feat: add ckawibow to cmu-courses #217

Merged
xboxbedrock merged 2 commits from ckawibow/governance:ckawibow-patch-1 into main 2026-09-19 21:32:17 +00:00
Member
No description provided.
feat: add ckawibow to cmu-courses.toml
All checks were successful
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 3m39s
CI / build (pull_request) Has been skipped
CI / check-1 (pull_request) Successful in 12m42s
CI / check (pull_request) Successful in 0s
bb20aa07c2
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
atlantis/plan 6/6 projects failed to plan.
kennel/build skipped: no services or static sites defined
atlantis/plan: google_play Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: discord Plan failed.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
9425de690e

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 8.01s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-courses_ch0_ckawibow will be created
+ resource "null_resource" "slack_cmu-courses_ch0_ckawibow" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0150RGAG1L"
          + "user"    = "U0BRZALGMT5"
        }
    }

  # null_resource.slack_cmu-study_ch0_annadavi will be destroyed
  # (because null_resource.slack_cmu-study_ch0_annadavi is not in configuration)
- resource "null_resource" "slack_cmu-study_ch0_annadavi" {
-       id       = "180323051751415711" -> null
-       triggers = {
          - "channel" = "C09KN6M3P53"
          - "user"    = "U07M24AKRGV"
        } -> null
    }

  # null_resource.slack_hub_annadavi will be destroyed
  # (because null_resource.slack_hub_annadavi is not in configuration)
- resource "null_resource" "slack_hub_annadavi" {
-       id       = "968070886944939043" -> null
-       triggers = {
          - "channel" = "C096TM8EMS8"
          - "user"    = "U07M24AKRGV"
        } -> null
    }

  # null_resource.slack_hub_ckawibow will be created
+ resource "null_resource" "slack_hub_ckawibow" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0BRZALGMT5"
        }
    }

  # null_resource.slack_leads_annadavi will be destroyed
  # (because null_resource.slack_leads_annadavi is not in configuration)
- resource "null_resource" "slack_leads_annadavi" {
-       id       = "4222181416230461163" -> null
-       triggers = {
          - "channel" = "C0AN1M0HDV0"
          - "user"    = "U07M24AKRGV"
        } -> null
    }

Plan: 2 to add, 0 to change, 3 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 3 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 8.05s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # googleplay_app_iam.courses_ckawibow will be created
+ resource "googleplay_app_iam" "courses_ckawibow" {
+       app_id               = "courses.cmu.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "ckawibow@andrew.cmu.edu"
    }

  # googleplay_user.ckawibow will be created
+ resource "googleplay_user" "ckawibow" {
+       email                = "ckawibow@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 2 to add, 0 to change, 0 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 8.03s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.ops_annadavi will be destroyed
  # (because google_cloud_identity_group_membership.ops_annadavi is not in configuration)
- resource "google_cloud_identity_group_membership" "ops_annadavi" {
-       create_time = "2026-09-19T00:09:12.648734Z" -> null
-       group       = "groups/03fwokq036zsp1f" -> null
-       id          = "groups/03fwokq036zsp1f/memberships/113975979101831473222" -> null
-       name        = "groups/03fwokq036zsp1f/memberships/113975979101831473222" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-19T00:09:12.648734Z" -> null

-       preferred_member_key {
-           id = "annadavi@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.tech_annadavi will be destroyed
  # (because google_cloud_identity_group_membership.tech_annadavi is not in configuration)
- resource "google_cloud_identity_group_membership" "tech_annadavi" {
-       create_time = "2026-09-19T00:09:12.648766Z" -> null
-       group       = "groups/04anzqyu1bqbv79" -> null
-       id          = "groups/04anzqyu1bqbv79/memberships/113975979101831473222" -> null
-       name        = "groups/04anzqyu1bqbv79/memberships/113975979101831473222" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-19T00:09:12.648766Z" -> null

-       preferred_member_key {
-           id = "annadavi@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.tech_ckawibow will be created
+ resource "google_cloud_identity_group_membership" "tech_ckawibow" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "ckawibow@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 2 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 8.01s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)
- destroy

OpenTofu will perform the following actions:

  # keycloak_group.project_cmu-study_study will be destroyed
  # (because keycloak_group.project_cmu-study_study is not in configuration)
- resource "keycloak_group" "project_cmu-study_study" {
-       attributes = {} -> null
-       id         = "f1b424fe-27f7-40ef-831f-9885f66bac3a" -> null
-       name       = "study" -> null
-       parent_id  = "9e6890c8-50a0-4b82-86b9-91791fffac25" -> null
-       path       = "/projects/study" -> null
-       realm_id   = "scottylabs" -> null
    }

  # keycloak_group.project_cmu-study_study_admins will be destroyed
  # (because keycloak_group.project_cmu-study_study_admins is not in configuration)
- resource "keycloak_group" "project_cmu-study_study_admins" {
-       attributes = {} -> null
-       id         = "a683866f-c161-4298-8d00-911d611e2895" -> null
-       name       = "admins" -> null
-       parent_id  = "f1b424fe-27f7-40ef-831f-9885f66bac3a" -> null
-       path       = "/projects/study/admins" -> null
-       realm_id   = "scottylabs" -> null
    }

  # keycloak_group_memberships.project_cmu-courses will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-courses" {
        id       = "scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77"
!       members  = [
+           "ckawibow",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-courses_courses will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-courses_courses" {
        id       = "scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406"
!       members  = [
+           "ckawibow",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-study_study will be destroyed
  # (because keycloak_group_memberships.project_cmu-study_study is not in configuration)
- resource "keycloak_group_memberships" "project_cmu-study_study" {
-       group_id = "f1b424fe-27f7-40ef-831f-9885f66bac3a" -> null
-       id       = "scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a" -> null
-       members  = [
-           "annadavi",
        ] -> null
-       realm_id = "scottylabs" -> null
    }

  # keycloak_group_memberships.project_cmu-study_study_admins will be destroyed
  # (because keycloak_group_memberships.project_cmu-study_study_admins is not in configuration)
- resource "keycloak_group_memberships" "project_cmu-study_study_admins" {
-       group_id = "a683866f-c161-4298-8d00-911d611e2895" -> null
-       id       = "scottylabs/group-memberships/a683866f-c161-4298-8d00-911d611e2895" -> null
-       members  = [
-           "annadavi",
        ] -> null
-       realm_id = "scottylabs" -> null
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
-           "annadavi",
+           "ckawibow",
            # (35 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_openid_client.study will be destroyed
  # (because keycloak_openid_client.study is not in configuration)
- resource "keycloak_openid_client" "study" {
-       access_type                                = "CONFIDENTIAL" -> null
-       always_display_in_console                  = false -> null
-       backchannel_logout_revoke_offline_sessions = false -> null
-       backchannel_logout_session_required        = true -> null
-       client_authenticator_type                  = "client-secret" -> null
-       client_id                                  = "study" -> null
-       client_secret                              = (sensitive value) -> null
-       client_secret_wo                           = (write-only attribute) -> null
-       consent_required                           = false -> null
-       direct_access_grants_enabled               = false -> null
-       display_on_consent_screen                  = false -> null
-       enabled                                    = true -> null
-       extra_config                               = {} -> null
-       frontchannel_logout_enabled                = false -> null
-       full_scope_allowed                         = true -> null
-       id                                         = "7622bc88-d377-4e9b-8995-86ec8e6f4638" -> null
-       implicit_flow_enabled                      = false -> null
-       import                                     = false -> null
-       name                                       = "study" -> null
-       oauth2_device_authorization_grant_enabled  = false -> null
-       oauth2_jwt_authorization_grant_enabled     = false -> null
-       realm_id                                   = "scottylabs" -> null
-       require_dpop_bound_tokens                  = false -> null
-       service_accounts_enabled                   = false -> null
-       standard_flow_enabled                      = true -> null
-       standard_token_exchange_enabled            = false -> null
-       use_refresh_tokens                         = true -> null
-       use_refresh_tokens_client_credentials      = false -> null
-       valid_post_logout_redirect_uris            = [] -> null
-       valid_redirect_uris                        = [
-           "https://oauth.scottylabs.org/oauth2/callback",
        ] -> null
-       web_origins                                = [
-           "+",
        ] -> null
    }

  # keycloak_openid_client.study_dev will be destroyed
  # (because keycloak_openid_client.study_dev is not in configuration)
- resource "keycloak_openid_client" "study_dev" {
-       access_type                                = "CONFIDENTIAL" -> null
-       always_display_in_console                  = false -> null
-       backchannel_logout_revoke_offline_sessions = false -> null
-       backchannel_logout_session_required        = true -> null
-       client_authenticator_type                  = "client-secret" -> null
-       client_id                                  = "study-dev" -> null
-       client_secret                              = (sensitive value) -> null
-       client_secret_wo                           = (write-only attribute) -> null
-       consent_required                           = false -> null
-       direct_access_grants_enabled               = false -> null
-       display_on_consent_screen                  = false -> null
-       enabled                                    = true -> null
-       extra_config                               = {} -> null
-       frontchannel_logout_enabled                = false -> null
-       full_scope_allowed                         = true -> null
-       id                                         = "1983432b-c99e-4470-9a59-1f8fdb9f8bef" -> null
-       implicit_flow_enabled                      = false -> null
-       import                                     = false -> null
-       name                                       = "study-dev" -> null
-       oauth2_device_authorization_grant_enabled  = false -> null
-       oauth2_jwt_authorization_grant_enabled     = false -> null
-       realm_id                                   = "scottylabs" -> null
-       require_dpop_bound_tokens                  = false -> null
-       service_accounts_enabled                   = false -> null
-       standard_flow_enabled                      = true -> null
-       standard_token_exchange_enabled            = false -> null
-       use_refresh_tokens                         = true -> null
-       use_refresh_tokens_client_credentials      = false -> null
-       valid_post_logout_redirect_uris            = [] -> null
-       valid_redirect_uris                        = [
-           "http://localhost:8090/oauth2/callback",
        ] -> null
-       web_origins                                = [
-           "+",
        ] -> null
    }

  # keycloak_openid_client.study_staging will be destroyed
  # (because keycloak_openid_client.study_staging is not in configuration)
- resource "keycloak_openid_client" "study_staging" {
-       access_type                                = "CONFIDENTIAL" -> null
-       always_display_in_console                  = false -> null
-       backchannel_logout_revoke_offline_sessions = false -> null
-       backchannel_logout_session_required        = true -> null
-       client_authenticator_type                  = "client-secret" -> null
-       client_id                                  = "study-staging" -> null
-       client_secret                              = (sensitive value) -> null
-       client_secret_wo                           = (write-only attribute) -> null
-       consent_required                           = false -> null
-       direct_access_grants_enabled               = false -> null
-       display_on_consent_screen                  = false -> null
-       enabled                                    = true -> null
-       extra_config                               = {} -> null
-       frontchannel_logout_enabled                = false -> null
-       full_scope_allowed                         = true -> null
-       id                                         = "18b84a0e-d9fa-4e99-9456-db6fba3358ff" -> null
-       implicit_flow_enabled                      = false -> null
-       import                                     = false -> null
-       name                                       = "study-staging" -> null
-       oauth2_device_authorization_grant_enabled  = false -> null
-       oauth2_jwt_authorization_grant_enabled     = false -> null
-       realm_id                                   = "scottylabs" -> null
-       require_dpop_bound_tokens                  = false -> null
-       service_accounts_enabled                   = false -> null
-       standard_flow_enabled                      = true -> null
-       standard_token_exchange_enabled            = false -> null
-       use_refresh_tokens                         = true -> null
-       use_refresh_tokens_client_credentials      = false -> null
-       valid_post_logout_redirect_uris            = [] -> null
-       valid_redirect_uris                        = [
-           "https://oauth.scottylabs.org/oauth2/callback",
        ] -> null
-       web_origins                                = [
-           "+",
        ] -> null
    }

  # keycloak_openid_group_membership_protocol_mapper.study will be destroyed
  # (because keycloak_openid_group_membership_protocol_mapper.study is not in configuration)
- resource "keycloak_openid_group_membership_protocol_mapper" "study" {
-       add_to_access_token        = true -> null
-       add_to_id_token            = true -> null
-       add_to_token_introspection = false -> null
-       add_to_userinfo            = true -> null
-       claim_name                 = "groups" -> null
-       client_id                  = "7622bc88-d377-4e9b-8995-86ec8e6f4638" -> null
-       full_path                  = true -> null
-       id                         = "54904eb2-e8ce-4ed7-9e50-565f4314ceeb" -> null
-       name                       = "groups" -> null
-       realm_id                   = "scottylabs" -> null
    }

  # keycloak_openid_group_membership_protocol_mapper.study_dev will be destroyed
  # (because keycloak_openid_group_membership_protocol_mapper.study_dev is not in configuration)
- resource "keycloak_openid_group_membership_protocol_mapper" "study_dev" {
-       add_to_access_token        = true -> null
-       add_to_id_token            = true -> null
-       add_to_token_introspection = false -> null
-       add_to_userinfo            = true -> null
-       claim_name                 = "groups" -> null
-       client_id                  = "1983432b-c99e-4470-9a59-1f8fdb9f8bef" -> null
-       full_path                  = true -> null
-       id                         = "d597e770-36f4-43bf-b9e3-04fa5228b29a" -> null
-       name                       = "groups" -> null
-       realm_id                   = "scottylabs" -> null
    }

  # keycloak_openid_group_membership_protocol_mapper.study_staging will be destroyed
  # (because keycloak_openid_group_membership_protocol_mapper.study_staging is not in configuration)
- resource "keycloak_openid_group_membership_protocol_mapper" "study_staging" {
-       add_to_access_token        = true -> null
-       add_to_id_token            = true -> null
-       add_to_token_introspection = false -> null
-       add_to_userinfo            = true -> null
-       claim_name                 = "groups" -> null
-       client_id                  = "18b84a0e-d9fa-4e99-9456-db6fba3358ff" -> null
-       full_path                  = true -> null
-       id                         = "7c989eb0-3f73-4692-b66c-4712c956f4ca" -> null
-       name                       = "groups" -> null
-       realm_id                   = "scottylabs" -> null
    }

  # vault_kv_secret_v2.study_dev_keycloak_realm will be destroyed
  # (because vault_kv_secret_v2.study_dev_keycloak_realm is not in configuration)
- resource "vault_kv_secret_v2" "study_dev_keycloak_realm" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/dev/KEYCLOAK_REALM" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.444714948Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/dev/KEYCLOAK_REALM" -> null
-       path                = "secret/data/secretspec/study/dev/KEYCLOAK_REALM" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_dev_keycloak_url will be destroyed
  # (because vault_kv_secret_v2.study_dev_keycloak_url is not in configuration)
- resource "vault_kv_secret_v2" "study_dev_keycloak_url" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/dev/KEYCLOAK_URL" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.442412478Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/dev/KEYCLOAK_URL" -> null
-       path                = "secret/data/secretspec/study/dev/KEYCLOAK_URL" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_dev_oauth_relay_url will be destroyed
  # (because vault_kv_secret_v2.study_dev_oauth_relay_url is not in configuration)
- resource "vault_kv_secret_v2" "study_dev_oauth_relay_url" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/dev/OAUTH_RELAY_URL" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.298044568Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/dev/OAUTH_RELAY_URL" -> null
-       path                = "secret/data/secretspec/study/dev/OAUTH_RELAY_URL" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_dev_oidc_client_id will be destroyed
  # (because vault_kv_secret_v2.study_dev_oidc_client_id is not in configuration)
- resource "vault_kv_secret_v2" "study_dev_oidc_client_id" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/dev/OIDC_CLIENT_ID" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.587736136Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/dev/OIDC_CLIENT_ID" -> null
-       path                = "secret/data/secretspec/study/dev/OIDC_CLIENT_ID" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_dev_oidc_client_secret will be destroyed
  # (because vault_kv_secret_v2.study_dev_oidc_client_secret is not in configuration)
- resource "vault_kv_secret_v2" "study_dev_oidc_client_secret" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/dev/OIDC_CLIENT_SECRET" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.585448593Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/dev/OIDC_CLIENT_SECRET" -> null
-       path                = "secret/data/secretspec/study/dev/OIDC_CLIENT_SECRET" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_dev_project_admin_group will be destroyed
  # (because vault_kv_secret_v2.study_dev_project_admin_group is not in configuration)
- resource "vault_kv_secret_v2" "study_dev_project_admin_group" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/dev/PROJECT_ADMIN_GROUP" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.293178762Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/dev/PROJECT_ADMIN_GROUP" -> null
-       path                = "secret/data/secretspec/study/dev/PROJECT_ADMIN_GROUP" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_dev_project_group will be destroyed
  # (because vault_kv_secret_v2.study_dev_project_group is not in configuration)
- resource "vault_kv_secret_v2" "study_dev_project_group" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/dev/PROJECT_GROUP" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.546407978Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/dev/PROJECT_GROUP" -> null
-       path                = "secret/data/secretspec/study/dev/PROJECT_GROUP" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_preview_keycloak_realm will be destroyed
  # (because vault_kv_secret_v2.study_preview_keycloak_realm is not in configuration)
- resource "vault_kv_secret_v2" "study_preview_keycloak_realm" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/preview/KEYCLOAK_REALM" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.298002244Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/preview/KEYCLOAK_REALM" -> null
-       path                = "secret/data/secretspec/study/preview/KEYCLOAK_REALM" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_preview_keycloak_url will be destroyed
  # (because vault_kv_secret_v2.study_preview_keycloak_url is not in configuration)
- resource "vault_kv_secret_v2" "study_preview_keycloak_url" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/preview/KEYCLOAK_URL" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.53837999Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/preview/KEYCLOAK_URL" -> null
-       path                = "secret/data/secretspec/study/preview/KEYCLOAK_URL" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_preview_oauth_relay_url will be destroyed
  # (because vault_kv_secret_v2.study_preview_oauth_relay_url is not in configuration)
- resource "vault_kv_secret_v2" "study_preview_oauth_relay_url" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/preview/OAUTH_RELAY_URL" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.438351034Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/preview/OAUTH_RELAY_URL" -> null
-       path                = "secret/data/secretspec/study/preview/OAUTH_RELAY_URL" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_preview_oidc_client_id will be destroyed
  # (because vault_kv_secret_v2.study_preview_oidc_client_id is not in configuration)
- resource "vault_kv_secret_v2" "study_preview_oidc_client_id" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/preview/OIDC_CLIENT_ID" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.711384417Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/preview/OIDC_CLIENT_ID" -> null
-       path                = "secret/data/secretspec/study/preview/OIDC_CLIENT_ID" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_preview_oidc_client_secret will be destroyed
  # (because vault_kv_secret_v2.study_preview_oidc_client_secret is not in configuration)
- resource "vault_kv_secret_v2" "study_preview_oidc_client_secret" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/preview/OIDC_CLIENT_SECRET" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.667580028Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/preview/OIDC_CLIENT_SECRET" -> null
-       path                = "secret/data/secretspec/study/preview/OIDC_CLIENT_SECRET" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_preview_project_admin_group will be destroyed
  # (because vault_kv_secret_v2.study_preview_project_admin_group is not in configuration)
- resource "vault_kv_secret_v2" "study_preview_project_admin_group" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/preview/PROJECT_ADMIN_GROUP" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.396806503Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/preview/PROJECT_ADMIN_GROUP" -> null
-       path                = "secret/data/secretspec/study/preview/PROJECT_ADMIN_GROUP" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_preview_project_group will be destroyed
  # (because vault_kv_secret_v2.study_preview_project_group is not in configuration)
- resource "vault_kv_secret_v2" "study_preview_project_group" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/preview/PROJECT_GROUP" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.551137659Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/preview/PROJECT_GROUP" -> null
-       path                = "secret/data/secretspec/study/preview/PROJECT_GROUP" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_prod_keycloak_realm will be destroyed
  # (because vault_kv_secret_v2.study_prod_keycloak_realm is not in configuration)
- resource "vault_kv_secret_v2" "study_prod_keycloak_realm" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/prod/KEYCLOAK_REALM" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.308833067Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/prod/KEYCLOAK_REALM" -> null
-       path                = "secret/data/secretspec/study/prod/KEYCLOAK_REALM" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_prod_keycloak_url will be destroyed
  # (because vault_kv_secret_v2.study_prod_keycloak_url is not in configuration)
- resource "vault_kv_secret_v2" "study_prod_keycloak_url" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/prod/KEYCLOAK_URL" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.396433099Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/prod/KEYCLOAK_URL" -> null
-       path                = "secret/data/secretspec/study/prod/KEYCLOAK_URL" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_prod_oauth_relay_url will be destroyed
  # (because vault_kv_secret_v2.study_prod_oauth_relay_url is not in configuration)
- resource "vault_kv_secret_v2" "study_prod_oauth_relay_url" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/prod/OAUTH_RELAY_URL" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.386052315Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/prod/OAUTH_RELAY_URL" -> null
-       path                = "secret/data/secretspec/study/prod/OAUTH_RELAY_URL" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_prod_oidc_client_id will be destroyed
  # (because vault_kv_secret_v2.study_prod_oidc_client_id is not in configuration)
- resource "vault_kv_secret_v2" "study_prod_oidc_client_id" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/prod/OIDC_CLIENT_ID" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.724442485Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/prod/OIDC_CLIENT_ID" -> null
-       path                = "secret/data/secretspec/study/prod/OIDC_CLIENT_ID" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_prod_oidc_client_secret will be destroyed
  # (because vault_kv_secret_v2.study_prod_oidc_client_secret is not in configuration)
- resource "vault_kv_secret_v2" "study_prod_oidc_client_secret" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/prod/OIDC_CLIENT_SECRET" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.592123604Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/prod/OIDC_CLIENT_SECRET" -> null
-       path                = "secret/data/secretspec/study/prod/OIDC_CLIENT_SECRET" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_prod_project_admin_group will be destroyed
  # (because vault_kv_secret_v2.study_prod_project_admin_group is not in configuration)
- resource "vault_kv_secret_v2" "study_prod_project_admin_group" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/prod/PROJECT_ADMIN_GROUP" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.440437759Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/prod/PROJECT_ADMIN_GROUP" -> null
-       path                = "secret/data/secretspec/study/prod/PROJECT_ADMIN_GROUP" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_prod_project_group will be destroyed
  # (because vault_kv_secret_v2.study_prod_project_group is not in configuration)
- resource "vault_kv_secret_v2" "study_prod_project_group" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/prod/PROJECT_GROUP" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.438313295Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/prod/PROJECT_GROUP" -> null
-       path                = "secret/data/secretspec/study/prod/PROJECT_GROUP" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_staging_keycloak_realm will be destroyed
  # (because vault_kv_secret_v2.study_staging_keycloak_realm is not in configuration)
- resource "vault_kv_secret_v2" "study_staging_keycloak_realm" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/staging/KEYCLOAK_REALM" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.428357393Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/staging/KEYCLOAK_REALM" -> null
-       path                = "secret/data/secretspec/study/staging/KEYCLOAK_REALM" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_staging_keycloak_url will be destroyed
  # (because vault_kv_secret_v2.study_staging_keycloak_url is not in configuration)
- resource "vault_kv_secret_v2" "study_staging_keycloak_url" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/staging/KEYCLOAK_URL" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.296273433Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/staging/KEYCLOAK_URL" -> null
-       path                = "secret/data/secretspec/study/staging/KEYCLOAK_URL" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_staging_oauth_relay_url will be destroyed
  # (because vault_kv_secret_v2.study_staging_oauth_relay_url is not in configuration)
- resource "vault_kv_secret_v2" "study_staging_oauth_relay_url" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/staging/OAUTH_RELAY_URL" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.440805344Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/staging/OAUTH_RELAY_URL" -> null
-       path                = "secret/data/secretspec/study/staging/OAUTH_RELAY_URL" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_staging_oidc_client_id will be destroyed
  # (because vault_kv_secret_v2.study_staging_oidc_client_id is not in configuration)
- resource "vault_kv_secret_v2" "study_staging_oidc_client_id" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/staging/OIDC_CLIENT_ID" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.672376407Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/staging/OIDC_CLIENT_ID" -> null
-       path                = "secret/data/secretspec/study/staging/OIDC_CLIENT_ID" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_staging_oidc_client_secret will be destroyed
  # (because vault_kv_secret_v2.study_staging_oidc_client_secret is not in configuration)
- resource "vault_kv_secret_v2" "study_staging_oidc_client_secret" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/staging/OIDC_CLIENT_SECRET" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.709411222Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/staging/OIDC_CLIENT_SECRET" -> null
-       path                = "secret/data/secretspec/study/staging/OIDC_CLIENT_SECRET" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_staging_project_admin_group will be destroyed
  # (because vault_kv_secret_v2.study_staging_project_admin_group is not in configuration)
- resource "vault_kv_secret_v2" "study_staging_project_admin_group" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/staging/PROJECT_ADMIN_GROUP" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.290931242Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/staging/PROJECT_ADMIN_GROUP" -> null
-       path                = "secret/data/secretspec/study/staging/PROJECT_ADMIN_GROUP" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

  # vault_kv_secret_v2.study_staging_project_group will be destroyed
  # (because vault_kv_secret_v2.study_staging_project_group is not in configuration)
- resource "vault_kv_secret_v2" "study_staging_project_group" {
-       data                = (sensitive value) -> null
-       data_json           = (sensitive value) -> null
-       data_json_wo        = (write-only attribute) -> null
-       delete_all_versions = false -> null
-       disable_read        = false -> null
-       id                  = "secret/data/secretspec/study/staging/PROJECT_GROUP" -> null
-       metadata            = {
          - "created_time"    = "2026-09-19T00:09:12.30080708Z"
          - "custom_metadata" = "null"
          - "deletion_time"   = ""
          - "destroyed"       = "false"
          - "version"         = "1"
        } -> null
-       mount               = "secret" -> null
-       name                = "secretspec/study/staging/PROJECT_GROUP" -> null
-       path                = "secret/data/secretspec/study/staging/PROJECT_GROUP" -> null

-       custom_metadata {
-           cas_required         = false -> null
-           data                 = {} -> null
-           delete_version_after = 0 -> null
-           max_versions         = 0 -> null
        }
    }

Plan: 0 to add, 3 to change, 38 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 3 to change, 38 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
   Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/217/default/crates/governance-schema)
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/217/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/217/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/217/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 7.98s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # discord_channel_permission.cmu-study_ch0 will be destroyed
  # (because discord_channel_permission.cmu-study_ch0 is not in configuration)
- resource "discord_channel_permission" "cmu-study_ch0" {
-       allow        = 1024 -> null
-       channel_id   = "1512270354943578132" -> null
-       deny         = 0 -> null
-       id           = "1512270354943578132:1550659990266581074:role" -> null
-       overwrite_id = "1550659990266581074" -> null
-       type         = "role" -> null
    }

  # discord_member_roles.annadavi will be destroyed
  # (because discord_member_roles.annadavi is not in configuration)
- resource "discord_member_roles" "annadavi" {
-       id        = "1353118209024331948:991766392590311455" -> null
-       server_id = "1353118209024331948" -> null
-       user_id   = "991766392590311455" -> null

-       role {
-           has_role = true -> null
-           role_id  = "1488787088118644848" -> null
        }
-       role {
-           has_role = true -> null
-           role_id  = "1519792320755662928" -> null
        }
-       role {
-           has_role = true -> null
-           role_id  = "1550659990266581074" -> null
        }
    }

  # discord_member_roles.ckawibow will be created
+ resource "discord_member_roles" "ckawibow" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "719730540323799112"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1500016758700904488"
        }
    }

  # discord_role.cmu-study will be destroyed
  # (because discord_role.cmu-study is not in configuration)
- resource "discord_role" "cmu-study" {
-       color       = 0 -> null
-       hoist       = false -> null
-       id          = "1550659990266581074" -> null
-       managed     = false -> null
-       mentionable = false -> null
-       name        = "CMU Study" -> null
-       permissions = 0 -> null
-       position    = 1 -> null
-       server_id   = "1353118209024331948" -> null
    }

Plan: 1 to add, 0 to change, 3 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 3 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 8.01s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # forgejo_repository.cmu-study_study will be destroyed
  # (because forgejo_repository.cmu-study_study is not in configuration)
- resource "forgejo_repository" "cmu-study_study" {
-       allow_manual_merge          = false -> null
-       allow_merge_commits         = true -> null
-       allow_rebase                = true -> null
-       allow_rebase_explicit       = true -> null
-       allow_squash_merge          = true -> null
-       archive_on_destroy          = false -> null
-       archived                    = false -> null
-       auto_init                   = true -> null
-       autodetect_manual_merge     = false -> null
-       clone_addr                  = "https://github.com/scottylabs-labrador/StudyStarter" -> null
-       clone_url                   = "https://git.cmu.dev/ScottyLabs/study.git" -> null
-       created_at                  = "2026-09-18T22:52:34Z" -> null
-       default_branch              = "main" -> null
-       default_merge_style         = "merge" -> null
-       description                 = "A platform for creating and joining study groups" -> null
-       empty                       = false -> null
-       fork                        = false -> null
-       forks_count                 = 0 -> null
-       full_name                   = "ScottyLabs/study" -> null
-       has_actions                 = true -> null
-       has_issues                  = true -> null
-       has_packages                = true -> null
-       has_projects                = true -> null
-       has_pull_requests           = true -> null
-       has_releases                = true -> null
-       has_wiki                    = true -> null
-       html_url                    = "https://git.cmu.dev/ScottyLabs/study" -> null
-       id                          = 93 -> null
-       ignore_whitespace_conflicts = false -> null
-       internal                    = false -> null
-       internal_tracker            = {
-           allow_only_contributors_to_track_time = true -> null
-           enable_issue_dependencies             = true -> null
-           enable_time_tracker                   = true -> null
        } -> null
-       labels                      = false -> null
-       lfs                         = false -> null
-       milestones                  = false -> null
-       mirror                      = false -> null
-       mirror_updated              = "0001-01-01T00:00:00Z" -> null
-       name                        = "study" -> null
-       open_issues_count           = 0 -> null
-       open_pr_counter             = 0 -> null
-       owner                       = "ScottyLabs" -> null
-       permissions                 = {
-           admin = true -> null
-           pull  = true -> null
-           push  = true -> null
        } -> null
-       private                     = false -> null
-       release_counter             = 0 -> null
-       size                        = 2234 -> null
-       ssh_url                     = "ssh://forgejo@git.cmu.dev/ScottyLabs/study.git" -> null
-       stars_count                 = 0 -> null
-       template                    = false -> null
-       trust_model                 = "default" -> null
-       updated_at                  = "2026-09-19T00:20:12Z" -> null
-       watchers_count              = 4 -> null
-       website                     = "https://cmustudy.com" -> null
    }

  # forgejo_team.cmu-study will be destroyed
  # (because forgejo_team.cmu-study is not in configuration)
- resource "forgejo_team" "cmu-study" {
-       can_create_org_repo       = false -> null
-       description               = "CMU Study" -> null
-       id                        = 43 -> null
-       includes_all_repositories = false -> null
-       name                      = "cmu-study" -> null
-       organization              = "ScottyLabs" -> null
-       organization_id           = 3 -> null
-       permission                = "write" -> null
-       units_map                 = {
          - "repo.code"   = "write"
          - "repo.issues" = "write"
          - "repo.pulls"  = "write"
        } -> null
    }

  # forgejo_team.study_leads will be destroyed
  # (because forgejo_team.study_leads is not in configuration)
- resource "forgejo_team" "study_leads" {
-       can_create_org_repo       = false -> null
-       description               = "CMU Study leads" -> null
-       id                        = 42 -> null
-       includes_all_repositories = false -> null
-       name                      = "study-leads" -> null
-       organization              = "ScottyLabs" -> null
-       organization_id           = 3 -> null
-       permission                = "admin" -> null
-       units_map                 = {
          - "repo.actions"    = "admin"
          - "repo.code"       = "admin"
          - "repo.ext_issues" = "admin"
          - "repo.ext_wiki"   = "admin"
          - "repo.issues"     = "admin"
          - "repo.packages"   = "admin"
          - "repo.projects"   = "admin"
          - "repo.pulls"      = "admin"
          - "repo.releases"   = "admin"
          - "repo.wiki"       = "admin"
        } -> null
    }

  # forgejo_team_member.cmu-courses_ckawibow will be created
+ resource "forgejo_team_member" "cmu-courses_ckawibow" {
+       team_id = 12
+       user    = "ckawibow"
    }

  # forgejo_team_member.cmu-study_annadavi will be destroyed
  # (because forgejo_team_member.cmu-study_annadavi is not in configuration)
- resource "forgejo_team_member" "cmu-study_annadavi" {
-       team_id = 43 -> null
-       user    = "annadavi" -> null
    }

  # forgejo_team_member.study_leads_annadavi will be destroyed
  # (because forgejo_team_member.study_leads_annadavi is not in configuration)
- resource "forgejo_team_member" "study_leads_annadavi" {
-       team_id = 42 -> null
-       user    = "annadavi" -> null
    }

  # forgejo_team_repository.cmu-study_study will be destroyed
  # (because forgejo_team_repository.cmu-study_study is not in configuration)
- resource "forgejo_team_repository" "cmu-study_study" {
-       owner      = "ScottyLabs" -> null
-       repository = "study" -> null
-       team_id    = 43 -> null
    }

  # forgejo_team_repository.study_leads_study will be destroyed
  # (because forgejo_team_repository.study_leads_study is not in configuration)
- resource "forgejo_team_repository" "study_leads_study" {
-       owner      = "ScottyLabs" -> null
-       repository = "study" -> null
-       team_id    = 42 -> null
    }

  # github_actions_repository_permissions.cmu-study_study_actions will be destroyed
  # (because github_actions_repository_permissions.cmu-study_study_actions is not in configuration)
- resource "github_actions_repository_permissions" "cmu-study_study_actions" {
-       enabled              = false -> null
-       id                   = "study" -> null
-       repository           = "study" -> null
-       sha_pinning_required = false -> null
    }

  # github_repository.cmu-study_study will be destroyed
  # (because github_repository.cmu-study_study is not in configuration)
- resource "github_repository" "cmu-study_study" {
-       allow_auto_merge                        = false -> null
-       allow_forking                           = true -> null
-       allow_merge_commit                      = true -> null
-       allow_rebase_merge                      = true -> null
-       allow_squash_merge                      = true -> null
-       allow_update_branch                     = false -> null
-       archived                                = false -> null
-       default_branch                          = "main" -> null
-       delete_branch_on_merge                  = false -> null
-       description                             = "READ-ONLY MIRROR" -> null
-       etag                                    = "W/\"8048e9b23088096330e4788fe625a163c3f48b66373b835592ddcc616d9edf1c\"" -> null
-       fork                                    = "false" -> null
-       full_name                               = "ScottyLabs/study" -> null
-       git_clone_url                           = "git://github.com/ScottyLabs/study.git" -> null
-       has_discussions                         = false -> null
-       has_downloads                           = false -> null
-       has_issues                              = false -> null
-       has_projects                            = false -> null
-       has_wiki                                = false -> null
-       homepage_url                            = "https://git.cmu.dev/ScottyLabs/study" -> null
-       html_url                                = "https://github.com/ScottyLabs/study" -> null
-       http_clone_url                          = "https://github.com/ScottyLabs/study.git" -> null
-       id                                      = "study" -> null
-       ignore_vulnerability_alerts_during_read = false -> null
-       is_template                             = false -> null
-       merge_commit_message                    = "PR_TITLE" -> null
-       merge_commit_title                      = "MERGE_MESSAGE" -> null
-       name                                    = "study" -> null
-       node_id                                 = "R_kgDOUgyHoQ" -> null
-       primary_language                        = "TypeScript" -> null
-       private                                 = false -> null
-       repo_id                                 = 1376552865 -> null
-       squash_merge_commit_message             = "COMMIT_MESSAGES" -> null
-       squash_merge_commit_title               = "COMMIT_OR_PR_TITLE" -> null
-       ssh_clone_url                           = "git@github.com:ScottyLabs/study.git" -> null
-       svn_url                                 = "https://github.com/ScottyLabs/study" -> null
-       topics                                  = [] -> null
-       visibility                              = "public" -> null
-       vulnerability_alerts                    = true -> null
-       web_commit_signoff_required             = false -> null

-       security_and_analysis {
-           secret_scanning {
-               status = "disabled" -> null
            }
-           secret_scanning_push_protection {
-               status = "disabled" -> null
            }
        }
    }

  # github_repository_deploy_key.cmu-study_study_mirror_deploy_key will be destroyed
  # (because github_repository_deploy_key.cmu-study_study_mirror_deploy_key is not in configuration)
- resource "github_repository_deploy_key" "cmu-study_study_mirror_deploy_key" {
-       etag       = "W/\"6040d9d76d3279913e7a31a43c0ae7c9a4ab7acb6b8b22024622cad92d49429e\"" -> null
-       id         = "study:163760029" -> null
-       key        = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFXVoILRqix/gcRujfj9pVK3gionHNuHbQSiuxz7DQ3G" -> null
-       read_only  = false -> null
-       repository = "study" -> null
-       title      = "Forgejo mirror" -> null
    }

  # github_repository_ruleset.cmu-study_study_ruleset will be destroyed
  # (because github_repository_ruleset.cmu-study_study_ruleset is not in configuration)
- resource "github_repository_ruleset" "cmu-study_study_ruleset" {
-       enforcement = "active" -> null
-       etag        = "\"908b9d0b9b6cc9b4ba22b9b004900021a51397da307a6f69815bd1f6067f176d\"" -> null
-       id          = "23686597" -> null
-       name        = "Default" -> null
-       node_id     = "RRS_lACqUmVwb3NpdG9yec5SDIehzgFpbcU" -> null
-       repository  = "study" -> null
-       ruleset_id  = 23686597 -> null
-       target      = "branch" -> null

-       bypass_actors {
-           actor_id    = 163760029 -> null
-           actor_type  = "DeployKey" -> null
-           bypass_mode = "always" -> null
        }

-       conditions {
-           ref_name {
-               exclude = [] -> null
-               include = [
-                   "~ALL",
                ] -> null
            }
        }

-       rules {
-           creation                      = true -> null
-           deletion                      = true -> null
-           non_fast_forward              = true -> null
-           required_linear_history       = false -> null
-           required_signatures           = false -> null
-           update                        = true -> null
-           update_allows_fetch_and_merge = false -> null

-           pull_request {
-               allowed_merge_methods             = [
-                   "merge",
-                   "squash",
-                   "rebase",
                ] -> null
-               dismiss_stale_reviews_on_push     = false -> null
-               require_code_owner_review         = false -> null
-               require_last_push_approval        = false -> null
-               required_approving_review_count   = 1 -> null
-               required_review_thread_resolution = false -> null
            }
        }
    }

  # github_team.cmu-study will be destroyed
  # (because github_team.cmu-study is not in configuration)
- resource "github_team" "cmu-study" {
-       create_default_maintainer = false -> null
-       etag                      = "W/\"d03a38c2a59aa76918965ee6a5b4ad5cc88a59aa6d642c7d6d8f3d5a75ee4f84\"" -> null
-       id                        = "19579842" -> null
-       members_count             = 1 -> null
-       name                      = "CMU Study" -> null
-       node_id                   = "T_kwDOABVDIs4BKsPC" -> null
-       notification_setting      = "notifications_enabled" -> null
-       privacy                   = "closed" -> null
-       slug                      = "cmu-study" -> null
    }

  # github_team_membership.cmu-courses_ckawibow will be created
+ resource "github_team_membership" "cmu-courses_ckawibow" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "3643907"
+       username = "chrkaw152"
    }

  # github_team_membership.cmu-study_annadavi will be destroyed
  # (because github_team_membership.cmu-study_annadavi is not in configuration)
- resource "github_team_membership" "cmu-study_annadavi" {
-       etag     = "W/\"e45bbdb4608cf5c568fb4c983e3458bdb42dc66d32fddf147208c8473c01e68a\"" -> null
-       id       = "19579842:annadavid1" -> null
-       role     = "maintainer" -> null
-       team_id  = "19579842" -> null
-       username = "annadavid1" -> null
    }

  # restapi_object.cmu-study_study_kennel_webhook will be destroyed
  # (because restapi_object.cmu-study_study_kennel_webhook is not in configuration)
- resource "restapi_object" "cmu-study_study_kennel_webhook" {
-       api_data        = {
          - "active"               = "true"
          - "authorization_header" = ""
          - "branch_filter"        = ""
          - "config"               = "map[content_type:json url:https://kennel.scottylabs.org/webhook]"
          - "content_type"         = "json"
          - "created_at"           = "2026-09-19T00:20:13Z"
          - "events"               = "[delete push pull_request pull_request_assign pull_request_label pull_request_milestone pull_request_comment pull_request_review_approved pull_request_review_rejected pull_request_review_comment pull_request_sync pull_request_review_request]"
          - "id"                   = "33"
          - "metadata"             = "<nil>"
          - "type"                 = "forgejo"
          - "updated_at"           = "2026-09-19T00:20:13Z"
          - "url"                  = "https://kennel.scottylabs.org/webhook"
        } -> null
-       api_response    = jsonencode(
            {
-               active               = true
-               authorization_header = ""
-               branch_filter        = ""
-               config               = {
-                   content_type = "json"
-                   url          = "https://kennel.scottylabs.org/webhook"
                }
-               content_type         = "json"
-               created_at           = "2026-09-19T00:20:13Z"
-               events               = [
-                   "delete",
-                   "push",
-                   "pull_request",
-                   "pull_request_assign",
-                   "pull_request_label",
-                   "pull_request_milestone",
-                   "pull_request_comment",
-                   "pull_request_review_approved",
-                   "pull_request_review_rejected",
-                   "pull_request_review_comment",
-                   "pull_request_sync",
-                   "pull_request_review_request",
                ]
-               id                   = 33
-               metadata             = null
-               type                 = "forgejo"
-               updated_at           = "2026-09-19T00:20:13Z"
-               url                  = "https://kennel.scottylabs.org/webhook"
            }
        ) -> null
-       create_response = jsonencode(
            {
-               active               = true
-               authorization_header = ""
-               branch_filter        = ""
-               config               = {
-                   content_type = "json"
-                   url          = "https://kennel.scottylabs.org/webhook"
                }
-               content_type         = "json"
-               created_at           = "2026-09-19T00:20:13Z"
-               events               = [
-                   "delete",
-                   "push",
-                   "pull_request",
-                   "pull_request_assign",
-                   "pull_request_label",
-                   "pull_request_milestone",
-                   "pull_request_comment",
-                   "pull_request_review_approved",
-                   "pull_request_review_rejected",
-                   "pull_request_review_comment",
-                   "pull_request_sync",
-                   "pull_request_review_request",
                ]
-               id                   = 33
-               metadata             = null
-               type                 = "forgejo"
-               updated_at           = "2026-09-19T00:20:13Z"
-               url                  = "https://kennel.scottylabs.org/webhook"
            }
        ) -> null
-       data            = (sensitive value) -> null
-       id              = "33" -> null
-       id_attribute    = "id" -> null
-       path            = "/api/v1/repos/ScottyLabs/study/hooks" -> null
-       update_method   = "PATCH" -> null
    }

  # restapi_object.cmu-study_study_push_mirror will be destroyed
  # (because restapi_object.cmu-study_study_push_mirror is not in configuration)
- resource "restapi_object" "cmu-study_study_push_mirror" {
-       api_data        = {
          - "branch_filter"  = ""
          - "created"        = "2026-09-19T00:20:13Z"
          - "interval"       = "8h0m0s"
          - "last_error"     = ""
          - "last_update"    = "2026-09-19T00:21:18Z"
          - "public_key"     = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFXVoILRqix/gcRujfj9pVK3gionHNuHbQSiuxz7DQ3G"
          - "remote_address" = "ssh://github.com/ScottyLabs/study.git"
          - "remote_name"    = "remote_mirror_Lo5Rz3Yz7Zp"
          - "repo_name"      = "study"
          - "sync_on_commit" = "true"
        } -> null
-       api_response    = jsonencode(
            {
-               branch_filter  = ""
-               created        = "2026-09-19T00:20:13Z"
-               interval       = "8h0m0s"
-               last_error     = ""
-               last_update    = "2026-09-19T00:21:18Z"
-               public_key     = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFXVoILRqix/gcRujfj9pVK3gionHNuHbQSiuxz7DQ3G"
-               remote_address = "ssh://github.com/ScottyLabs/study.git"
-               remote_name    = "remote_mirror_Lo5Rz3Yz7Zp"
-               repo_name      = "study"
-               sync_on_commit = true
            }
        ) -> null
-       create_response = jsonencode(
            {
-               branch_filter  = ""
-               created        = "2026-09-19T00:20:13Z"
-               interval       = "8h0m0s"
-               last_error     = ""
-               last_update    = "1970-01-01T00:00:00Z"
-               public_key     = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFXVoILRqix/gcRujfj9pVK3gionHNuHbQSiuxz7DQ3G"
-               remote_address = "ssh://github.com/ScottyLabs/study.git"
-               remote_name    = "remote_mirror_Lo5Rz3Yz7Zp"
-               repo_name      = "study"
-               sync_on_commit = true
            }
        ) -> null
-       data            = jsonencode(
            {
-               interval       = "8h0m0s"
-               remote_address = "git@github.com:ScottyLabs/study.git"
-               sync_on_commit = true
-               use_ssh        = true
            }
        ) -> null
-       id              = "remote_mirror_Lo5Rz3Yz7Zp" -> null
-       id_attribute    = "remote_name" -> null
-       path            = "/api/v1/repos/ScottyLabs/study/push_mirrors" -> null
    }

Plan: 2 to add, 0 to change, 15 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 0 to change, 15 to destroy.


Plan Summary

6 projects, 6 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 8.01s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # null_resource.slack_cmu-courses_ch0_ckawibow will be created + resource "null_resource" "slack_cmu-courses_ch0_ckawibow" { + id = (known after apply) + triggers = { + "channel" = "C0150RGAG1L" + "user" = "U0BRZALGMT5" } } # null_resource.slack_cmu-study_ch0_annadavi will be destroyed # (because null_resource.slack_cmu-study_ch0_annadavi is not in configuration) - resource "null_resource" "slack_cmu-study_ch0_annadavi" { - id = "180323051751415711" -> null - triggers = { - "channel" = "C09KN6M3P53" - "user" = "U07M24AKRGV" } -> null } # null_resource.slack_hub_annadavi will be destroyed # (because null_resource.slack_hub_annadavi is not in configuration) - resource "null_resource" "slack_hub_annadavi" { - id = "968070886944939043" -> null - triggers = { - "channel" = "C096TM8EMS8" - "user" = "U07M24AKRGV" } -> null } # null_resource.slack_hub_ckawibow will be created + resource "null_resource" "slack_hub_ckawibow" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0BRZALGMT5" } } # null_resource.slack_leads_annadavi will be destroyed # (because null_resource.slack_leads_annadavi is not in configuration) - resource "null_resource" "slack_leads_annadavi" { - id = "4222181416230461163" -> null - triggers = { - "channel" = "C0AN1M0HDV0" - "user" = "U07M24AKRGV" } -> null } Plan: 2 to add, 0 to change, 3 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 3 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 8.05s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # googleplay_app_iam.courses_ckawibow will be created + resource "googleplay_app_iam" "courses_ckawibow" { + app_id = "courses.cmu.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "ckawibow@andrew.cmu.edu" } # googleplay_user.ckawibow will be created + resource "googleplay_user" "ckawibow" { + email = "ckawibow@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 8.03s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # google_cloud_identity_group_membership.ops_annadavi will be destroyed # (because google_cloud_identity_group_membership.ops_annadavi is not in configuration) - resource "google_cloud_identity_group_membership" "ops_annadavi" { - create_time = "2026-09-19T00:09:12.648734Z" -> null - group = "groups/03fwokq036zsp1f" -> null - id = "groups/03fwokq036zsp1f/memberships/113975979101831473222" -> null - name = "groups/03fwokq036zsp1f/memberships/113975979101831473222" -> null - type = "USER" -> null - update_time = "2026-09-19T00:09:12.648734Z" -> null - preferred_member_key { - id = "annadavi@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.tech_annadavi will be destroyed # (because google_cloud_identity_group_membership.tech_annadavi is not in configuration) - resource "google_cloud_identity_group_membership" "tech_annadavi" { - create_time = "2026-09-19T00:09:12.648766Z" -> null - group = "groups/04anzqyu1bqbv79" -> null - id = "groups/04anzqyu1bqbv79/memberships/113975979101831473222" -> null - name = "groups/04anzqyu1bqbv79/memberships/113975979101831473222" -> null - type = "USER" -> null - update_time = "2026-09-19T00:09:12.648766Z" -> null - preferred_member_key { - id = "annadavi@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.tech_ckawibow will be created + resource "google_cloud_identity_group_membership" "tech_ckawibow" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "ckawibow@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 2 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 8.01s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) - destroy OpenTofu will perform the following actions: # keycloak_group.project_cmu-study_study will be destroyed # (because keycloak_group.project_cmu-study_study is not in configuration) - resource "keycloak_group" "project_cmu-study_study" { - attributes = {} -> null - id = "f1b424fe-27f7-40ef-831f-9885f66bac3a" -> null - name = "study" -> null - parent_id = "9e6890c8-50a0-4b82-86b9-91791fffac25" -> null - path = "/projects/study" -> null - realm_id = "scottylabs" -> null } # keycloak_group.project_cmu-study_study_admins will be destroyed # (because keycloak_group.project_cmu-study_study_admins is not in configuration) - resource "keycloak_group" "project_cmu-study_study_admins" { - attributes = {} -> null - id = "a683866f-c161-4298-8d00-911d611e2895" -> null - name = "admins" -> null - parent_id = "f1b424fe-27f7-40ef-831f-9885f66bac3a" -> null - path = "/projects/study/admins" -> null - realm_id = "scottylabs" -> null } # keycloak_group_memberships.project_cmu-courses will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-courses" { id = "scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77" ! members = [ + "ckawibow", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-courses_courses will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-courses_courses" { id = "scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406" ! members = [ + "ckawibow", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-study_study will be destroyed # (because keycloak_group_memberships.project_cmu-study_study is not in configuration) - resource "keycloak_group_memberships" "project_cmu-study_study" { - group_id = "f1b424fe-27f7-40ef-831f-9885f66bac3a" -> null - id = "scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a" -> null - members = [ - "annadavi", ] -> null - realm_id = "scottylabs" -> null } # keycloak_group_memberships.project_cmu-study_study_admins will be destroyed # (because keycloak_group_memberships.project_cmu-study_study_admins is not in configuration) - resource "keycloak_group_memberships" "project_cmu-study_study_admins" { - group_id = "a683866f-c161-4298-8d00-911d611e2895" -> null - id = "scottylabs/group-memberships/a683866f-c161-4298-8d00-911d611e2895" -> null - members = [ - "annadavi", ] -> null - realm_id = "scottylabs" -> null } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ - "annadavi", + "ckawibow", # (35 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_openid_client.study will be destroyed # (because keycloak_openid_client.study is not in configuration) - resource "keycloak_openid_client" "study" { - access_type = "CONFIDENTIAL" -> null - always_display_in_console = false -> null - backchannel_logout_revoke_offline_sessions = false -> null - backchannel_logout_session_required = true -> null - client_authenticator_type = "client-secret" -> null - client_id = "study" -> null - client_secret = (sensitive value) -> null - client_secret_wo = (write-only attribute) -> null - consent_required = false -> null - direct_access_grants_enabled = false -> null - display_on_consent_screen = false -> null - enabled = true -> null - extra_config = {} -> null - frontchannel_logout_enabled = false -> null - full_scope_allowed = true -> null - id = "7622bc88-d377-4e9b-8995-86ec8e6f4638" -> null - implicit_flow_enabled = false -> null - import = false -> null - name = "study" -> null - oauth2_device_authorization_grant_enabled = false -> null - oauth2_jwt_authorization_grant_enabled = false -> null - realm_id = "scottylabs" -> null - require_dpop_bound_tokens = false -> null - service_accounts_enabled = false -> null - standard_flow_enabled = true -> null - standard_token_exchange_enabled = false -> null - use_refresh_tokens = true -> null - use_refresh_tokens_client_credentials = false -> null - valid_post_logout_redirect_uris = [] -> null - valid_redirect_uris = [ - "https://oauth.scottylabs.org/oauth2/callback", ] -> null - web_origins = [ - "+", ] -> null } # keycloak_openid_client.study_dev will be destroyed # (because keycloak_openid_client.study_dev is not in configuration) - resource "keycloak_openid_client" "study_dev" { - access_type = "CONFIDENTIAL" -> null - always_display_in_console = false -> null - backchannel_logout_revoke_offline_sessions = false -> null - backchannel_logout_session_required = true -> null - client_authenticator_type = "client-secret" -> null - client_id = "study-dev" -> null - client_secret = (sensitive value) -> null - client_secret_wo = (write-only attribute) -> null - consent_required = false -> null - direct_access_grants_enabled = false -> null - display_on_consent_screen = false -> null - enabled = true -> null - extra_config = {} -> null - frontchannel_logout_enabled = false -> null - full_scope_allowed = true -> null - id = "1983432b-c99e-4470-9a59-1f8fdb9f8bef" -> null - implicit_flow_enabled = false -> null - import = false -> null - name = "study-dev" -> null - oauth2_device_authorization_grant_enabled = false -> null - oauth2_jwt_authorization_grant_enabled = false -> null - realm_id = "scottylabs" -> null - require_dpop_bound_tokens = false -> null - service_accounts_enabled = false -> null - standard_flow_enabled = true -> null - standard_token_exchange_enabled = false -> null - use_refresh_tokens = true -> null - use_refresh_tokens_client_credentials = false -> null - valid_post_logout_redirect_uris = [] -> null - valid_redirect_uris = [ - "http://localhost:8090/oauth2/callback", ] -> null - web_origins = [ - "+", ] -> null } # keycloak_openid_client.study_staging will be destroyed # (because keycloak_openid_client.study_staging is not in configuration) - resource "keycloak_openid_client" "study_staging" { - access_type = "CONFIDENTIAL" -> null - always_display_in_console = false -> null - backchannel_logout_revoke_offline_sessions = false -> null - backchannel_logout_session_required = true -> null - client_authenticator_type = "client-secret" -> null - client_id = "study-staging" -> null - client_secret = (sensitive value) -> null - client_secret_wo = (write-only attribute) -> null - consent_required = false -> null - direct_access_grants_enabled = false -> null - display_on_consent_screen = false -> null - enabled = true -> null - extra_config = {} -> null - frontchannel_logout_enabled = false -> null - full_scope_allowed = true -> null - id = "18b84a0e-d9fa-4e99-9456-db6fba3358ff" -> null - implicit_flow_enabled = false -> null - import = false -> null - name = "study-staging" -> null - oauth2_device_authorization_grant_enabled = false -> null - oauth2_jwt_authorization_grant_enabled = false -> null - realm_id = "scottylabs" -> null - require_dpop_bound_tokens = false -> null - service_accounts_enabled = false -> null - standard_flow_enabled = true -> null - standard_token_exchange_enabled = false -> null - use_refresh_tokens = true -> null - use_refresh_tokens_client_credentials = false -> null - valid_post_logout_redirect_uris = [] -> null - valid_redirect_uris = [ - "https://oauth.scottylabs.org/oauth2/callback", ] -> null - web_origins = [ - "+", ] -> null } # keycloak_openid_group_membership_protocol_mapper.study will be destroyed # (because keycloak_openid_group_membership_protocol_mapper.study is not in configuration) - resource "keycloak_openid_group_membership_protocol_mapper" "study" { - add_to_access_token = true -> null - add_to_id_token = true -> null - add_to_token_introspection = false -> null - add_to_userinfo = true -> null - claim_name = "groups" -> null - client_id = "7622bc88-d377-4e9b-8995-86ec8e6f4638" -> null - full_path = true -> null - id = "54904eb2-e8ce-4ed7-9e50-565f4314ceeb" -> null - name = "groups" -> null - realm_id = "scottylabs" -> null } # keycloak_openid_group_membership_protocol_mapper.study_dev will be destroyed # (because keycloak_openid_group_membership_protocol_mapper.study_dev is not in configuration) - resource "keycloak_openid_group_membership_protocol_mapper" "study_dev" { - add_to_access_token = true -> null - add_to_id_token = true -> null - add_to_token_introspection = false -> null - add_to_userinfo = true -> null - claim_name = "groups" -> null - client_id = "1983432b-c99e-4470-9a59-1f8fdb9f8bef" -> null - full_path = true -> null - id = "d597e770-36f4-43bf-b9e3-04fa5228b29a" -> null - name = "groups" -> null - realm_id = "scottylabs" -> null } # keycloak_openid_group_membership_protocol_mapper.study_staging will be destroyed # (because keycloak_openid_group_membership_protocol_mapper.study_staging is not in configuration) - resource "keycloak_openid_group_membership_protocol_mapper" "study_staging" { - add_to_access_token = true -> null - add_to_id_token = true -> null - add_to_token_introspection = false -> null - add_to_userinfo = true -> null - claim_name = "groups" -> null - client_id = "18b84a0e-d9fa-4e99-9456-db6fba3358ff" -> null - full_path = true -> null - id = "7c989eb0-3f73-4692-b66c-4712c956f4ca" -> null - name = "groups" -> null - realm_id = "scottylabs" -> null } # vault_kv_secret_v2.study_dev_keycloak_realm will be destroyed # (because vault_kv_secret_v2.study_dev_keycloak_realm is not in configuration) - resource "vault_kv_secret_v2" "study_dev_keycloak_realm" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/dev/KEYCLOAK_REALM" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.444714948Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/dev/KEYCLOAK_REALM" -> null - path = "secret/data/secretspec/study/dev/KEYCLOAK_REALM" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_dev_keycloak_url will be destroyed # (because vault_kv_secret_v2.study_dev_keycloak_url is not in configuration) - resource "vault_kv_secret_v2" "study_dev_keycloak_url" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/dev/KEYCLOAK_URL" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.442412478Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/dev/KEYCLOAK_URL" -> null - path = "secret/data/secretspec/study/dev/KEYCLOAK_URL" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_dev_oauth_relay_url will be destroyed # (because vault_kv_secret_v2.study_dev_oauth_relay_url is not in configuration) - resource "vault_kv_secret_v2" "study_dev_oauth_relay_url" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/dev/OAUTH_RELAY_URL" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.298044568Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/dev/OAUTH_RELAY_URL" -> null - path = "secret/data/secretspec/study/dev/OAUTH_RELAY_URL" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_dev_oidc_client_id will be destroyed # (because vault_kv_secret_v2.study_dev_oidc_client_id is not in configuration) - resource "vault_kv_secret_v2" "study_dev_oidc_client_id" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/dev/OIDC_CLIENT_ID" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.587736136Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/dev/OIDC_CLIENT_ID" -> null - path = "secret/data/secretspec/study/dev/OIDC_CLIENT_ID" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_dev_oidc_client_secret will be destroyed # (because vault_kv_secret_v2.study_dev_oidc_client_secret is not in configuration) - resource "vault_kv_secret_v2" "study_dev_oidc_client_secret" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/dev/OIDC_CLIENT_SECRET" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.585448593Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/dev/OIDC_CLIENT_SECRET" -> null - path = "secret/data/secretspec/study/dev/OIDC_CLIENT_SECRET" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_dev_project_admin_group will be destroyed # (because vault_kv_secret_v2.study_dev_project_admin_group is not in configuration) - resource "vault_kv_secret_v2" "study_dev_project_admin_group" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/dev/PROJECT_ADMIN_GROUP" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.293178762Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/dev/PROJECT_ADMIN_GROUP" -> null - path = "secret/data/secretspec/study/dev/PROJECT_ADMIN_GROUP" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_dev_project_group will be destroyed # (because vault_kv_secret_v2.study_dev_project_group is not in configuration) - resource "vault_kv_secret_v2" "study_dev_project_group" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/dev/PROJECT_GROUP" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.546407978Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/dev/PROJECT_GROUP" -> null - path = "secret/data/secretspec/study/dev/PROJECT_GROUP" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_preview_keycloak_realm will be destroyed # (because vault_kv_secret_v2.study_preview_keycloak_realm is not in configuration) - resource "vault_kv_secret_v2" "study_preview_keycloak_realm" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/preview/KEYCLOAK_REALM" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.298002244Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/preview/KEYCLOAK_REALM" -> null - path = "secret/data/secretspec/study/preview/KEYCLOAK_REALM" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_preview_keycloak_url will be destroyed # (because vault_kv_secret_v2.study_preview_keycloak_url is not in configuration) - resource "vault_kv_secret_v2" "study_preview_keycloak_url" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/preview/KEYCLOAK_URL" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.53837999Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/preview/KEYCLOAK_URL" -> null - path = "secret/data/secretspec/study/preview/KEYCLOAK_URL" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_preview_oauth_relay_url will be destroyed # (because vault_kv_secret_v2.study_preview_oauth_relay_url is not in configuration) - resource "vault_kv_secret_v2" "study_preview_oauth_relay_url" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/preview/OAUTH_RELAY_URL" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.438351034Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/preview/OAUTH_RELAY_URL" -> null - path = "secret/data/secretspec/study/preview/OAUTH_RELAY_URL" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_preview_oidc_client_id will be destroyed # (because vault_kv_secret_v2.study_preview_oidc_client_id is not in configuration) - resource "vault_kv_secret_v2" "study_preview_oidc_client_id" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/preview/OIDC_CLIENT_ID" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.711384417Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/preview/OIDC_CLIENT_ID" -> null - path = "secret/data/secretspec/study/preview/OIDC_CLIENT_ID" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_preview_oidc_client_secret will be destroyed # (because vault_kv_secret_v2.study_preview_oidc_client_secret is not in configuration) - resource "vault_kv_secret_v2" "study_preview_oidc_client_secret" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/preview/OIDC_CLIENT_SECRET" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.667580028Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/preview/OIDC_CLIENT_SECRET" -> null - path = "secret/data/secretspec/study/preview/OIDC_CLIENT_SECRET" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_preview_project_admin_group will be destroyed # (because vault_kv_secret_v2.study_preview_project_admin_group is not in configuration) - resource "vault_kv_secret_v2" "study_preview_project_admin_group" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/preview/PROJECT_ADMIN_GROUP" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.396806503Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/preview/PROJECT_ADMIN_GROUP" -> null - path = "secret/data/secretspec/study/preview/PROJECT_ADMIN_GROUP" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_preview_project_group will be destroyed # (because vault_kv_secret_v2.study_preview_project_group is not in configuration) - resource "vault_kv_secret_v2" "study_preview_project_group" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/preview/PROJECT_GROUP" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.551137659Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/preview/PROJECT_GROUP" -> null - path = "secret/data/secretspec/study/preview/PROJECT_GROUP" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_prod_keycloak_realm will be destroyed # (because vault_kv_secret_v2.study_prod_keycloak_realm is not in configuration) - resource "vault_kv_secret_v2" "study_prod_keycloak_realm" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/prod/KEYCLOAK_REALM" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.308833067Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/prod/KEYCLOAK_REALM" -> null - path = "secret/data/secretspec/study/prod/KEYCLOAK_REALM" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_prod_keycloak_url will be destroyed # (because vault_kv_secret_v2.study_prod_keycloak_url is not in configuration) - resource "vault_kv_secret_v2" "study_prod_keycloak_url" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/prod/KEYCLOAK_URL" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.396433099Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/prod/KEYCLOAK_URL" -> null - path = "secret/data/secretspec/study/prod/KEYCLOAK_URL" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_prod_oauth_relay_url will be destroyed # (because vault_kv_secret_v2.study_prod_oauth_relay_url is not in configuration) - resource "vault_kv_secret_v2" "study_prod_oauth_relay_url" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/prod/OAUTH_RELAY_URL" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.386052315Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/prod/OAUTH_RELAY_URL" -> null - path = "secret/data/secretspec/study/prod/OAUTH_RELAY_URL" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_prod_oidc_client_id will be destroyed # (because vault_kv_secret_v2.study_prod_oidc_client_id is not in configuration) - resource "vault_kv_secret_v2" "study_prod_oidc_client_id" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/prod/OIDC_CLIENT_ID" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.724442485Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/prod/OIDC_CLIENT_ID" -> null - path = "secret/data/secretspec/study/prod/OIDC_CLIENT_ID" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_prod_oidc_client_secret will be destroyed # (because vault_kv_secret_v2.study_prod_oidc_client_secret is not in configuration) - resource "vault_kv_secret_v2" "study_prod_oidc_client_secret" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/prod/OIDC_CLIENT_SECRET" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.592123604Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/prod/OIDC_CLIENT_SECRET" -> null - path = "secret/data/secretspec/study/prod/OIDC_CLIENT_SECRET" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_prod_project_admin_group will be destroyed # (because vault_kv_secret_v2.study_prod_project_admin_group is not in configuration) - resource "vault_kv_secret_v2" "study_prod_project_admin_group" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/prod/PROJECT_ADMIN_GROUP" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.440437759Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/prod/PROJECT_ADMIN_GROUP" -> null - path = "secret/data/secretspec/study/prod/PROJECT_ADMIN_GROUP" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_prod_project_group will be destroyed # (because vault_kv_secret_v2.study_prod_project_group is not in configuration) - resource "vault_kv_secret_v2" "study_prod_project_group" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/prod/PROJECT_GROUP" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.438313295Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/prod/PROJECT_GROUP" -> null - path = "secret/data/secretspec/study/prod/PROJECT_GROUP" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_staging_keycloak_realm will be destroyed # (because vault_kv_secret_v2.study_staging_keycloak_realm is not in configuration) - resource "vault_kv_secret_v2" "study_staging_keycloak_realm" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/staging/KEYCLOAK_REALM" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.428357393Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/staging/KEYCLOAK_REALM" -> null - path = "secret/data/secretspec/study/staging/KEYCLOAK_REALM" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_staging_keycloak_url will be destroyed # (because vault_kv_secret_v2.study_staging_keycloak_url is not in configuration) - resource "vault_kv_secret_v2" "study_staging_keycloak_url" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/staging/KEYCLOAK_URL" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.296273433Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/staging/KEYCLOAK_URL" -> null - path = "secret/data/secretspec/study/staging/KEYCLOAK_URL" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_staging_oauth_relay_url will be destroyed # (because vault_kv_secret_v2.study_staging_oauth_relay_url is not in configuration) - resource "vault_kv_secret_v2" "study_staging_oauth_relay_url" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/staging/OAUTH_RELAY_URL" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.440805344Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/staging/OAUTH_RELAY_URL" -> null - path = "secret/data/secretspec/study/staging/OAUTH_RELAY_URL" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_staging_oidc_client_id will be destroyed # (because vault_kv_secret_v2.study_staging_oidc_client_id is not in configuration) - resource "vault_kv_secret_v2" "study_staging_oidc_client_id" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/staging/OIDC_CLIENT_ID" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.672376407Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/staging/OIDC_CLIENT_ID" -> null - path = "secret/data/secretspec/study/staging/OIDC_CLIENT_ID" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_staging_oidc_client_secret will be destroyed # (because vault_kv_secret_v2.study_staging_oidc_client_secret is not in configuration) - resource "vault_kv_secret_v2" "study_staging_oidc_client_secret" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/staging/OIDC_CLIENT_SECRET" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.709411222Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/staging/OIDC_CLIENT_SECRET" -> null - path = "secret/data/secretspec/study/staging/OIDC_CLIENT_SECRET" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_staging_project_admin_group will be destroyed # (because vault_kv_secret_v2.study_staging_project_admin_group is not in configuration) - resource "vault_kv_secret_v2" "study_staging_project_admin_group" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/staging/PROJECT_ADMIN_GROUP" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.290931242Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/staging/PROJECT_ADMIN_GROUP" -> null - path = "secret/data/secretspec/study/staging/PROJECT_ADMIN_GROUP" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } # vault_kv_secret_v2.study_staging_project_group will be destroyed # (because vault_kv_secret_v2.study_staging_project_group is not in configuration) - resource "vault_kv_secret_v2" "study_staging_project_group" { - data = (sensitive value) -> null - data_json = (sensitive value) -> null - data_json_wo = (write-only attribute) -> null - delete_all_versions = false -> null - disable_read = false -> null - id = "secret/data/secretspec/study/staging/PROJECT_GROUP" -> null - metadata = { - "created_time" = "2026-09-19T00:09:12.30080708Z" - "custom_metadata" = "null" - "deletion_time" = "" - "destroyed" = "false" - "version" = "1" } -> null - mount = "secret" -> null - name = "secretspec/study/staging/PROJECT_GROUP" -> null - path = "secret/data/secretspec/study/staging/PROJECT_GROUP" -> null - custom_metadata { - cas_required = false -> null - data = {} -> null - delete_version_after = 0 -> null - max_versions = 0 -> null } } Plan: 0 to add, 3 to change, 38 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 3 to change, 38 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/217/default/crates/governance-schema) Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/217/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/217/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/217/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 7.98s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # discord_channel_permission.cmu-study_ch0 will be destroyed # (because discord_channel_permission.cmu-study_ch0 is not in configuration) - resource "discord_channel_permission" "cmu-study_ch0" { - allow = 1024 -> null - channel_id = "1512270354943578132" -> null - deny = 0 -> null - id = "1512270354943578132:1550659990266581074:role" -> null - overwrite_id = "1550659990266581074" -> null - type = "role" -> null } # discord_member_roles.annadavi will be destroyed # (because discord_member_roles.annadavi is not in configuration) - resource "discord_member_roles" "annadavi" { - id = "1353118209024331948:991766392590311455" -> null - server_id = "1353118209024331948" -> null - user_id = "991766392590311455" -> null - role { - has_role = true -> null - role_id = "1488787088118644848" -> null } - role { - has_role = true -> null - role_id = "1519792320755662928" -> null } - role { - has_role = true -> null - role_id = "1550659990266581074" -> null } } # discord_member_roles.ckawibow will be created + resource "discord_member_roles" "ckawibow" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "719730540323799112" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1500016758700904488" } } # discord_role.cmu-study will be destroyed # (because discord_role.cmu-study is not in configuration) - resource "discord_role" "cmu-study" { - color = 0 -> null - hoist = false -> null - id = "1550659990266581074" -> null - managed = false -> null - mentionable = false -> null - name = "CMU Study" -> null - permissions = 0 -> null - position = 1 -> null - server_id = "1353118209024331948" -> null } Plan: 1 to add, 0 to change, 3 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 3 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 8.01s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # forgejo_repository.cmu-study_study will be destroyed # (because forgejo_repository.cmu-study_study is not in configuration) - resource "forgejo_repository" "cmu-study_study" { - allow_manual_merge = false -> null - allow_merge_commits = true -> null - allow_rebase = true -> null - allow_rebase_explicit = true -> null - allow_squash_merge = true -> null - archive_on_destroy = false -> null - archived = false -> null - auto_init = true -> null - autodetect_manual_merge = false -> null - clone_addr = "https://github.com/scottylabs-labrador/StudyStarter" -> null - clone_url = "https://git.cmu.dev/ScottyLabs/study.git" -> null - created_at = "2026-09-18T22:52:34Z" -> null - default_branch = "main" -> null - default_merge_style = "merge" -> null - description = "A platform for creating and joining study groups" -> null - empty = false -> null - fork = false -> null - forks_count = 0 -> null - full_name = "ScottyLabs/study" -> null - has_actions = true -> null - has_issues = true -> null - has_packages = true -> null - has_projects = true -> null - has_pull_requests = true -> null - has_releases = true -> null - has_wiki = true -> null - html_url = "https://git.cmu.dev/ScottyLabs/study" -> null - id = 93 -> null - ignore_whitespace_conflicts = false -> null - internal = false -> null - internal_tracker = { - allow_only_contributors_to_track_time = true -> null - enable_issue_dependencies = true -> null - enable_time_tracker = true -> null } -> null - labels = false -> null - lfs = false -> null - milestones = false -> null - mirror = false -> null - mirror_updated = "0001-01-01T00:00:00Z" -> null - name = "study" -> null - open_issues_count = 0 -> null - open_pr_counter = 0 -> null - owner = "ScottyLabs" -> null - permissions = { - admin = true -> null - pull = true -> null - push = true -> null } -> null - private = false -> null - release_counter = 0 -> null - size = 2234 -> null - ssh_url = "ssh://forgejo@git.cmu.dev/ScottyLabs/study.git" -> null - stars_count = 0 -> null - template = false -> null - trust_model = "default" -> null - updated_at = "2026-09-19T00:20:12Z" -> null - watchers_count = 4 -> null - website = "https://cmustudy.com" -> null } # forgejo_team.cmu-study will be destroyed # (because forgejo_team.cmu-study is not in configuration) - resource "forgejo_team" "cmu-study" { - can_create_org_repo = false -> null - description = "CMU Study" -> null - id = 43 -> null - includes_all_repositories = false -> null - name = "cmu-study" -> null - organization = "ScottyLabs" -> null - organization_id = 3 -> null - permission = "write" -> null - units_map = { - "repo.code" = "write" - "repo.issues" = "write" - "repo.pulls" = "write" } -> null } # forgejo_team.study_leads will be destroyed # (because forgejo_team.study_leads is not in configuration) - resource "forgejo_team" "study_leads" { - can_create_org_repo = false -> null - description = "CMU Study leads" -> null - id = 42 -> null - includes_all_repositories = false -> null - name = "study-leads" -> null - organization = "ScottyLabs" -> null - organization_id = 3 -> null - permission = "admin" -> null - units_map = { - "repo.actions" = "admin" - "repo.code" = "admin" - "repo.ext_issues" = "admin" - "repo.ext_wiki" = "admin" - "repo.issues" = "admin" - "repo.packages" = "admin" - "repo.projects" = "admin" - "repo.pulls" = "admin" - "repo.releases" = "admin" - "repo.wiki" = "admin" } -> null } # forgejo_team_member.cmu-courses_ckawibow will be created + resource "forgejo_team_member" "cmu-courses_ckawibow" { + team_id = 12 + user = "ckawibow" } # forgejo_team_member.cmu-study_annadavi will be destroyed # (because forgejo_team_member.cmu-study_annadavi is not in configuration) - resource "forgejo_team_member" "cmu-study_annadavi" { - team_id = 43 -> null - user = "annadavi" -> null } # forgejo_team_member.study_leads_annadavi will be destroyed # (because forgejo_team_member.study_leads_annadavi is not in configuration) - resource "forgejo_team_member" "study_leads_annadavi" { - team_id = 42 -> null - user = "annadavi" -> null } # forgejo_team_repository.cmu-study_study will be destroyed # (because forgejo_team_repository.cmu-study_study is not in configuration) - resource "forgejo_team_repository" "cmu-study_study" { - owner = "ScottyLabs" -> null - repository = "study" -> null - team_id = 43 -> null } # forgejo_team_repository.study_leads_study will be destroyed # (because forgejo_team_repository.study_leads_study is not in configuration) - resource "forgejo_team_repository" "study_leads_study" { - owner = "ScottyLabs" -> null - repository = "study" -> null - team_id = 42 -> null } # github_actions_repository_permissions.cmu-study_study_actions will be destroyed # (because github_actions_repository_permissions.cmu-study_study_actions is not in configuration) - resource "github_actions_repository_permissions" "cmu-study_study_actions" { - enabled = false -> null - id = "study" -> null - repository = "study" -> null - sha_pinning_required = false -> null } # github_repository.cmu-study_study will be destroyed # (because github_repository.cmu-study_study is not in configuration) - resource "github_repository" "cmu-study_study" { - allow_auto_merge = false -> null - allow_forking = true -> null - allow_merge_commit = true -> null - allow_rebase_merge = true -> null - allow_squash_merge = true -> null - allow_update_branch = false -> null - archived = false -> null - default_branch = "main" -> null - delete_branch_on_merge = false -> null - description = "READ-ONLY MIRROR" -> null - etag = "W/\"8048e9b23088096330e4788fe625a163c3f48b66373b835592ddcc616d9edf1c\"" -> null - fork = "false" -> null - full_name = "ScottyLabs/study" -> null - git_clone_url = "git://github.com/ScottyLabs/study.git" -> null - has_discussions = false -> null - has_downloads = false -> null - has_issues = false -> null - has_projects = false -> null - has_wiki = false -> null - homepage_url = "https://git.cmu.dev/ScottyLabs/study" -> null - html_url = "https://github.com/ScottyLabs/study" -> null - http_clone_url = "https://github.com/ScottyLabs/study.git" -> null - id = "study" -> null - ignore_vulnerability_alerts_during_read = false -> null - is_template = false -> null - merge_commit_message = "PR_TITLE" -> null - merge_commit_title = "MERGE_MESSAGE" -> null - name = "study" -> null - node_id = "R_kgDOUgyHoQ" -> null - primary_language = "TypeScript" -> null - private = false -> null - repo_id = 1376552865 -> null - squash_merge_commit_message = "COMMIT_MESSAGES" -> null - squash_merge_commit_title = "COMMIT_OR_PR_TITLE" -> null - ssh_clone_url = "git@github.com:ScottyLabs/study.git" -> null - svn_url = "https://github.com/ScottyLabs/study" -> null - topics = [] -> null - visibility = "public" -> null - vulnerability_alerts = true -> null - web_commit_signoff_required = false -> null - security_and_analysis { - secret_scanning { - status = "disabled" -> null } - secret_scanning_push_protection { - status = "disabled" -> null } } } # github_repository_deploy_key.cmu-study_study_mirror_deploy_key will be destroyed # (because github_repository_deploy_key.cmu-study_study_mirror_deploy_key is not in configuration) - resource "github_repository_deploy_key" "cmu-study_study_mirror_deploy_key" { - etag = "W/\"6040d9d76d3279913e7a31a43c0ae7c9a4ab7acb6b8b22024622cad92d49429e\"" -> null - id = "study:163760029" -> null - key = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFXVoILRqix/gcRujfj9pVK3gionHNuHbQSiuxz7DQ3G" -> null - read_only = false -> null - repository = "study" -> null - title = "Forgejo mirror" -> null } # github_repository_ruleset.cmu-study_study_ruleset will be destroyed # (because github_repository_ruleset.cmu-study_study_ruleset is not in configuration) - resource "github_repository_ruleset" "cmu-study_study_ruleset" { - enforcement = "active" -> null - etag = "\"908b9d0b9b6cc9b4ba22b9b004900021a51397da307a6f69815bd1f6067f176d\"" -> null - id = "23686597" -> null - name = "Default" -> null - node_id = "RRS_lACqUmVwb3NpdG9yec5SDIehzgFpbcU" -> null - repository = "study" -> null - ruleset_id = 23686597 -> null - target = "branch" -> null - bypass_actors { - actor_id = 163760029 -> null - actor_type = "DeployKey" -> null - bypass_mode = "always" -> null } - conditions { - ref_name { - exclude = [] -> null - include = [ - "~ALL", ] -> null } } - rules { - creation = true -> null - deletion = true -> null - non_fast_forward = true -> null - required_linear_history = false -> null - required_signatures = false -> null - update = true -> null - update_allows_fetch_and_merge = false -> null - pull_request { - allowed_merge_methods = [ - "merge", - "squash", - "rebase", ] -> null - dismiss_stale_reviews_on_push = false -> null - require_code_owner_review = false -> null - require_last_push_approval = false -> null - required_approving_review_count = 1 -> null - required_review_thread_resolution = false -> null } } } # github_team.cmu-study will be destroyed # (because github_team.cmu-study is not in configuration) - resource "github_team" "cmu-study" { - create_default_maintainer = false -> null - etag = "W/\"d03a38c2a59aa76918965ee6a5b4ad5cc88a59aa6d642c7d6d8f3d5a75ee4f84\"" -> null - id = "19579842" -> null - members_count = 1 -> null - name = "CMU Study" -> null - node_id = "T_kwDOABVDIs4BKsPC" -> null - notification_setting = "notifications_enabled" -> null - privacy = "closed" -> null - slug = "cmu-study" -> null } # github_team_membership.cmu-courses_ckawibow will be created + resource "github_team_membership" "cmu-courses_ckawibow" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "3643907" + username = "chrkaw152" } # github_team_membership.cmu-study_annadavi will be destroyed # (because github_team_membership.cmu-study_annadavi is not in configuration) - resource "github_team_membership" "cmu-study_annadavi" { - etag = "W/\"e45bbdb4608cf5c568fb4c983e3458bdb42dc66d32fddf147208c8473c01e68a\"" -> null - id = "19579842:annadavid1" -> null - role = "maintainer" -> null - team_id = "19579842" -> null - username = "annadavid1" -> null } # restapi_object.cmu-study_study_kennel_webhook will be destroyed # (because restapi_object.cmu-study_study_kennel_webhook is not in configuration) - resource "restapi_object" "cmu-study_study_kennel_webhook" { - api_data = { - "active" = "true" - "authorization_header" = "" - "branch_filter" = "" - "config" = "map[content_type:json url:https://kennel.scottylabs.org/webhook]" - "content_type" = "json" - "created_at" = "2026-09-19T00:20:13Z" - "events" = "[delete push pull_request pull_request_assign pull_request_label pull_request_milestone pull_request_comment pull_request_review_approved pull_request_review_rejected pull_request_review_comment pull_request_sync pull_request_review_request]" - "id" = "33" - "metadata" = "<nil>" - "type" = "forgejo" - "updated_at" = "2026-09-19T00:20:13Z" - "url" = "https://kennel.scottylabs.org/webhook" } -> null - api_response = jsonencode( { - active = true - authorization_header = "" - branch_filter = "" - config = { - content_type = "json" - url = "https://kennel.scottylabs.org/webhook" } - content_type = "json" - created_at = "2026-09-19T00:20:13Z" - events = [ - "delete", - "push", - "pull_request", - "pull_request_assign", - "pull_request_label", - "pull_request_milestone", - "pull_request_comment", - "pull_request_review_approved", - "pull_request_review_rejected", - "pull_request_review_comment", - "pull_request_sync", - "pull_request_review_request", ] - id = 33 - metadata = null - type = "forgejo" - updated_at = "2026-09-19T00:20:13Z" - url = "https://kennel.scottylabs.org/webhook" } ) -> null - create_response = jsonencode( { - active = true - authorization_header = "" - branch_filter = "" - config = { - content_type = "json" - url = "https://kennel.scottylabs.org/webhook" } - content_type = "json" - created_at = "2026-09-19T00:20:13Z" - events = [ - "delete", - "push", - "pull_request", - "pull_request_assign", - "pull_request_label", - "pull_request_milestone", - "pull_request_comment", - "pull_request_review_approved", - "pull_request_review_rejected", - "pull_request_review_comment", - "pull_request_sync", - "pull_request_review_request", ] - id = 33 - metadata = null - type = "forgejo" - updated_at = "2026-09-19T00:20:13Z" - url = "https://kennel.scottylabs.org/webhook" } ) -> null - data = (sensitive value) -> null - id = "33" -> null - id_attribute = "id" -> null - path = "/api/v1/repos/ScottyLabs/study/hooks" -> null - update_method = "PATCH" -> null } # restapi_object.cmu-study_study_push_mirror will be destroyed # (because restapi_object.cmu-study_study_push_mirror is not in configuration) - resource "restapi_object" "cmu-study_study_push_mirror" { - api_data = { - "branch_filter" = "" - "created" = "2026-09-19T00:20:13Z" - "interval" = "8h0m0s" - "last_error" = "" - "last_update" = "2026-09-19T00:21:18Z" - "public_key" = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFXVoILRqix/gcRujfj9pVK3gionHNuHbQSiuxz7DQ3G" - "remote_address" = "ssh://github.com/ScottyLabs/study.git" - "remote_name" = "remote_mirror_Lo5Rz3Yz7Zp" - "repo_name" = "study" - "sync_on_commit" = "true" } -> null - api_response = jsonencode( { - branch_filter = "" - created = "2026-09-19T00:20:13Z" - interval = "8h0m0s" - last_error = "" - last_update = "2026-09-19T00:21:18Z" - public_key = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFXVoILRqix/gcRujfj9pVK3gionHNuHbQSiuxz7DQ3G" - remote_address = "ssh://github.com/ScottyLabs/study.git" - remote_name = "remote_mirror_Lo5Rz3Yz7Zp" - repo_name = "study" - sync_on_commit = true } ) -> null - create_response = jsonencode( { - branch_filter = "" - created = "2026-09-19T00:20:13Z" - interval = "8h0m0s" - last_error = "" - last_update = "1970-01-01T00:00:00Z" - public_key = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFXVoILRqix/gcRujfj9pVK3gionHNuHbQSiuxz7DQ3G" - remote_address = "ssh://github.com/ScottyLabs/study.git" - remote_name = "remote_mirror_Lo5Rz3Yz7Zp" - repo_name = "study" - sync_on_commit = true } ) -> null - data = jsonencode( { - interval = "8h0m0s" - remote_address = "git@github.com:ScottyLabs/study.git" - sync_on_commit = true - use_ssh = true } ) -> null - id = "remote_mirror_Lo5Rz3Yz7Zp" -> null - id_attribute = "remote_name" -> null - path = "/api/v1/repos/ScottyLabs/study/push_mirrors" -> null } Plan: 2 to add, 0 to change, 15 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 0 to change, 15 to destroy. --- ### Plan Summary 6 projects, 6 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis unlock

atlantis unlock

All Atlantis locks for this PR have been unlocked and plans discarded

All Atlantis locks for this PR have been unlocked and plans discarded
etashjha scheduled this pull request to auto merge when all checks succeed 2026-09-19 18:49:20 +00:00
etashjha canceled auto merging this pull request when all checks succeed 2026-09-19 18:53:38 +00:00
etashjha approved these changes 2026-09-19 18:54:18 +00:00
Member

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_groups dir: tofu/google_groups workspace: default
  3. project: google_play dir: tofu/google_play workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: discord dir: tofu/discord workspace: default

1. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis unlock

atlantis unlock

All Atlantis locks for this PR have been unlocked and plans discarded

All Atlantis locks for this PR have been unlocked and plans discarded
xboxbedrock force-pushed ckawibow-patch-1 from 9425de690e
Some checks failed
atlantis/plan 6/6 projects failed to plan.
kennel/build skipped: no services or static sites defined
atlantis/plan: google_play Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: discord Plan failed.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
to c1a2d5fd7b
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: git Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan 6/6 projects failed to plan.
kennel/build skipped: no services or static sites defined
2026-09-19 21:06:59 +00:00
Compare

Ran Plan for 6 projects:

  1. project: discord dir: tofu/discord workspace: default
  2. project: git dir: tofu/git workspace: default
  3. project: keycloak dir: tofu/keycloak workspace: default
  4. project: slack dir: tofu/slack workspace: default
  5. project: google_play dir: tofu/google_play workspace: default
  6. project: google_groups dir: tofu/google_groups workspace: default

1. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` --- ### 1. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
xboxbedrock force-pushed ckawibow-patch-1 from c1a2d5fd7b
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: git Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan 6/6 projects failed to plan.
kennel/build skipped: no services or static sites defined
to c4c93978f0
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: google_play Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan 6/6 projects failed to plan.
2026-09-19 21:14:12 +00:00
Compare

Ran Plan for 6 projects:

  1. project: discord dir: tofu/discord workspace: default
  2. project: keycloak dir: tofu/keycloak workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: google_play dir: tofu/google_play workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: slack dir: tofu/slack workspace: default

1. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #256. To continue, delete the lock from #256 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #256. To continue, delete the lock from #256 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #256. To continue, delete the lock from #256 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #256. To continue, delete the lock from #256 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #256. To continue, delete the lock from #256 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #256. To continue, delete the lock from #256 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` --- ### 1. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #256. To continue, delete the lock from #256 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #256. To continue, delete the lock from #256 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #256. To continue, delete the lock from #256 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #256. To continue, delete the lock from #256 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #256. To continue, delete the lock from #256 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #256. To continue, delete the lock from #256 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: slack dir: tofu/slack workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: keycloak dir: tofu/keycloak workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #261. To continue, delete the lock from #261 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #261. To continue, delete the lock from #261 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #261. To continue, delete the lock from #261 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #261. To continue, delete the lock from #261 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #261. To continue, delete the lock from #261 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #261. To continue, delete the lock from #261 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #261. To continue, delete the lock from #261 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #261. To continue, delete the lock from #261 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #261. To continue, delete the lock from #261 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #261. To continue, delete the lock from #261 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #261. To continue, delete the lock from #261 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #261. To continue, delete the lock from #261 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
xboxbedrock force-pushed ckawibow-patch-1 from c4c93978f0
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: google_play Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan 6/6 projects failed to plan.
to 882ed7749e
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_play Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 3 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
2026-09-19 21:28:44 +00:00
Compare

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: google_groups dir: tofu/google_groups workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #263. To continue, delete the lock from #263 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #263. To continue, delete the lock from #263 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #263. To continue, delete the lock from #263 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #263. To continue, delete the lock from #263 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #263. To continue, delete the lock from #263 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #263. To continue, delete the lock from #263 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #263. To continue, delete the lock from #263 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #263. To continue, delete the lock from #263 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #263. To continue, delete the lock from #263 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #263. To continue, delete the lock from #263 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #263. To continue, delete the lock from #263 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #263. To continue, delete the lock from #263 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.19s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-courses_ch0_ckawibow will be created
+ resource "null_resource" "slack_cmu-courses_ch0_ckawibow" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0150RGAG1L"
          + "user"    = "U0BRZALGMT5"
        }
    }

  # null_resource.slack_hub_ckawibow will be created
+ resource "null_resource" "slack_hub_ckawibow" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0BRZALGMT5"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 0 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.22s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # googleplay_app_iam.courses_ckawibow will be created
+ resource "googleplay_app_iam" "courses_ckawibow" {
+       app_id               = "courses.cmu.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "ckawibow@andrew.cmu.edu"
    }

  # googleplay_user.ckawibow will be created
+ resource "googleplay_user" "ckawibow" {
+       email                = "ckawibow@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 2 to add, 0 to change, 0 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.27s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_ckawibow will be created
+ resource "google_cloud_identity_group_membership" "tech_ckawibow" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "ckawibow@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-courses will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-courses" {
        id       = "scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77"
!       members  = [
+           "ckawibow",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-courses_courses will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-courses_courses" {
        id       = "scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406"
!       members  = [
+           "ckawibow",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "ckawibow",
            # (43 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 3 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 3 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.26s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.ckawibow will be created
+ resource "discord_member_roles" "ckawibow" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "719730540323799112"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1500016758700904488"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.18s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # forgejo_team_member.cmu-courses_ckawibow will be created
+ resource "forgejo_team_member" "cmu-courses_ckawibow" {
+       team_id = 12
+       user    = "ckawibow"
    }

  # github_team_membership.cmu-courses_ckawibow will be created
+ resource "github_team_membership" "cmu-courses_ckawibow" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "3643907"
+       username = "chrkaw152"
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 0 to change, 0 to destroy.


Plan Summary

6 projects, 6 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.19s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_cmu-courses_ch0_ckawibow will be created + resource "null_resource" "slack_cmu-courses_ch0_ckawibow" { + id = (known after apply) + triggers = { + "channel" = "C0150RGAG1L" + "user" = "U0BRZALGMT5" } } # null_resource.slack_hub_ckawibow will be created + resource "null_resource" "slack_hub_ckawibow" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0BRZALGMT5" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.22s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # googleplay_app_iam.courses_ckawibow will be created + resource "googleplay_app_iam" "courses_ckawibow" { + app_id = "courses.cmu.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "ckawibow@andrew.cmu.edu" } # googleplay_user.ckawibow will be created + resource "googleplay_user" "ckawibow" { + email = "ckawibow@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.27s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_ckawibow will be created + resource "google_cloud_identity_group_membership" "tech_ckawibow" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "ckawibow@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-courses will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-courses" { id = "scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77" ! members = [ + "ckawibow", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-courses_courses will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-courses_courses" { id = "scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406" ! members = [ + "ckawibow", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "ckawibow", # (43 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 3 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 3 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.26s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.ckawibow will be created + resource "discord_member_roles" "ckawibow" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "719730540323799112" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1500016758700904488" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.18s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # forgejo_team_member.cmu-courses_ckawibow will be created + resource "forgejo_team_member" "cmu-courses_ckawibow" { + team_id = 12 + user = "ckawibow" } # github_team_membership.cmu-courses_ckawibow will be created + resource "github_team_membership" "cmu-courses_ckawibow" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "3643907" + username = "chrkaw152" } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### Plan Summary 6 projects, 6 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis apply

atlantis apply

Ran Apply for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: google_play dir: tofu/google_play workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: keycloak dir: tofu/keycloak workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.44s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_cmu-courses_ch0_ckawibow: Creating...
null_resource.slack_cmu-courses_ch0_ckawibow: Provisioning with 'local-exec'...
null_resource.slack_cmu-courses_ch0_ckawibow (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0150RGAG1L --user U0BRZALGMT5"]
null_resource.slack_hub_ckawibow: Creating...
null_resource.slack_hub_ckawibow: Provisioning with 'local-exec'...
null_resource.slack_hub_ckawibow (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0BRZALGMT5"]
null_resource.slack_hub_ckawibow: Creation complete after 0s [id=6750382842144089636]
null_resource.slack_cmu-courses_ch0_ckawibow: Creation complete after 0s [id=6586365389488197591]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

2. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.33s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.ckawibow: Creating...
discord_member_roles.ckawibow: Creation complete after 0s [id=1353118209024331948:719730540323799112]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

3. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.33s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
googleplay_user.ckawibow: Creating...
googleplay_user.ckawibow: Creation complete after 1s [id=ckawibow@andrew.cmu.edu]
googleplay_app_iam.courses_ckawibow: Creating...
googleplay_app_iam.courses_ckawibow: Creation complete after 1s

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.13s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_ckawibow: Creating...
google_cloud_identity_group_membership.tech_ckawibow: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/114421018781322230816]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.20s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
github_team_membership.cmu-courses_ckawibow: Creating...
forgejo_team_member.cmu-courses_ckawibow: Creating...
forgejo_team_member.cmu-courses_ckawibow: Creation complete after 0s
github_team_membership.cmu-courses_ckawibow: Creation complete after 1s [id=3643907:chrkaw152]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

6. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.44s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cmu-courses: Modifying... [id=scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77]
keycloak_group_memberships.project_cmu-courses_courses: Modifying... [id=scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_cmu-courses_courses: Modifications complete after 2s [id=scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406]
keycloak_group_memberships.project_cmu-courses: Modifications complete after 2s [id=scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77]
keycloak_group_memberships.projects: Modifications complete after 2s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 3 changed, 0 destroyed.

Apply Summary

6 projects, 6 successful, 0 failed, 0 errored

Ran Apply for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.44s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_cmu-courses_ch0_ckawibow: Creating... null_resource.slack_cmu-courses_ch0_ckawibow: Provisioning with 'local-exec'... null_resource.slack_cmu-courses_ch0_ckawibow (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0150RGAG1L --user U0BRZALGMT5"] null_resource.slack_hub_ckawibow: Creating... null_resource.slack_hub_ckawibow: Provisioning with 'local-exec'... null_resource.slack_hub_ckawibow (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0BRZALGMT5"] null_resource.slack_hub_ckawibow: Creation complete after 0s [id=6750382842144089636] null_resource.slack_cmu-courses_ch0_ckawibow: Creation complete after 0s [id=6586365389488197591] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.33s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.ckawibow: Creating... discord_member_roles.ckawibow: Creation complete after 0s [id=1353118209024331948:719730540323799112] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.33s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ googleplay_user.ckawibow: Creating... googleplay_user.ckawibow: Creation complete after 1s [id=ckawibow@andrew.cmu.edu] googleplay_app_iam.courses_ckawibow: Creating... googleplay_app_iam.courses_ckawibow: Creation complete after 1s Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.13s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_ckawibow: Creating... google_cloud_identity_group_membership.tech_ckawibow: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/114421018781322230816] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.20s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ github_team_membership.cmu-courses_ckawibow: Creating... forgejo_team_member.cmu-courses_ckawibow: Creating... forgejo_team_member.cmu-courses_ckawibow: Creation complete after 0s github_team_membership.cmu-courses_ckawibow: Creation complete after 1s [id=3643907:chrkaw152] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 6. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.44s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cmu-courses: Modifying... [id=scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77] keycloak_group_memberships.project_cmu-courses_courses: Modifying... [id=scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_cmu-courses_courses: Modifications complete after 2s [id=scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406] keycloak_group_memberships.project_cmu-courses: Modifications complete after 2s [id=scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77] keycloak_group_memberships.projects: Modifications complete after 2s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 3 changed, 0 destroyed. ``` </details> --- ### Apply Summary 6 projects, 6 successful, 0 failed, 0 errored

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
Sign in to join this conversation.
No description provided.