feat: add andyj to cmu-study #225

Merged
xboxbedrock merged 2 commits from andyj/governance:add-name into main 2026-09-19 21:12:58 +00:00
Member
No description provided.
feat: add andyj to cmu-study
All checks were successful
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 2m54s
CI / build (pull_request) Has been skipped
CI / check-1 (pull_request) Successful in 13m44s
CI / check (pull_request) Successful in 0s
9c0eb4420f
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: slack Plan: 2 to add, 0 to change, 10 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 4 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 8 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 1 to change, 4 to destroy.
atlantis/plan: git Plan: 2 to add, 0 to change, 28 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
b07dd3f370

Ran Plan for 6 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: git dir: tofu/git workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: google_groups dir: tofu/google_groups workspace: default
  6. project: discord dir: tofu/discord workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
annadavi approved these changes 2026-09-19 19:22:42 +00:00
Member

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.29s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-study_ch0_andyj will be created
+ resource "null_resource" "slack_cmu-study_ch0_andyj" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C09KN6M3P53"
          + "user"    = "U0C2066T7T9"
        }
    }

  # null_resource.slack_devops_ch0_acchs will be destroyed
  # (because null_resource.slack_devops_ch0_acchs is not in configuration)
- resource "null_resource" "slack_devops_ch0_acchs" {
-       id       = "5494655407757332457" -> null
-       triggers = {
          - "channel" = "C08K3Q77ZQF"
          - "user"    = "U0BVBNX8YGZ"
        } -> null
    }

  # null_resource.slack_devops_ch0_aszhou will be destroyed
  # (because null_resource.slack_devops_ch0_aszhou is not in configuration)
- resource "null_resource" "slack_devops_ch0_aszhou" {
-       id       = "3525995290209722166" -> null
-       triggers = {
          - "channel" = "C08K3Q77ZQF"
          - "user"    = "U09CQ04V4BT"
        } -> null
    }

  # null_resource.slack_devops_ch0_indigo0445 will be destroyed
  # (because null_resource.slack_devops_ch0_indigo0445 is not in configuration)
- resource "null_resource" "slack_devops_ch0_indigo0445" {
-       id       = "166910966562392614" -> null
-       triggers = {
          - "channel" = "C08K3Q77ZQF"
          - "user"    = "U09C4RTR36F"
        } -> null
    }

  # null_resource.slack_devops_ch1_acchs will be destroyed
  # (because null_resource.slack_devops_ch1_acchs is not in configuration)
- resource "null_resource" "slack_devops_ch1_acchs" {
-       id       = "1105780819490615240" -> null
-       triggers = {
          - "channel" = "C0BNV84NUTC"
          - "user"    = "U0BVBNX8YGZ"
        } -> null
    }

  # null_resource.slack_devops_ch1_aszhou will be destroyed
  # (because null_resource.slack_devops_ch1_aszhou is not in configuration)
- resource "null_resource" "slack_devops_ch1_aszhou" {
-       id       = "4874526058879346740" -> null
-       triggers = {
          - "channel" = "C0BNV84NUTC"
          - "user"    = "U09CQ04V4BT"
        } -> null
    }

  # null_resource.slack_devops_ch1_indigo0445 will be destroyed
  # (because null_resource.slack_devops_ch1_indigo0445 is not in configuration)
- resource "null_resource" "slack_devops_ch1_indigo0445" {
-       id       = "7985592316574563315" -> null
-       triggers = {
          - "channel" = "C0BNV84NUTC"
          - "user"    = "U09C4RTR36F"
        } -> null
    }

  # null_resource.slack_hub_andyj will be created
+ resource "null_resource" "slack_hub_andyj" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0C2066T7T9"
        }
    }

  # null_resource.slack_hub_aszhou will be destroyed
  # (because null_resource.slack_hub_aszhou is not in configuration)
- resource "null_resource" "slack_hub_aszhou" {
-       id       = "2233394787102757076" -> null
-       triggers = {
          - "channel" = "C096TM8EMS8"
          - "user"    = "U09CQ04V4BT"
        } -> null
    }

  # null_resource.slack_hub_drishtis will be destroyed
  # (because null_resource.slack_hub_drishtis is not in configuration)
- resource "null_resource" "slack_hub_drishtis" {
-       id       = "465737434924418541" -> null
-       triggers = {
          - "channel" = "C096TM8EMS8"
          - "user"    = "U0BTH04N9SP"
        } -> null
    }

  # null_resource.slack_hub_indigo0445 will be destroyed
  # (because null_resource.slack_hub_indigo0445 is not in configuration)
- resource "null_resource" "slack_hub_indigo0445" {
-       id       = "5455653694878708973" -> null
-       triggers = {
          - "channel" = "C096TM8EMS8"
          - "user"    = "U09C4RTR36F"
        } -> null
    }

  # null_resource.slack_hub_xindiy will be destroyed
  # (because null_resource.slack_hub_xindiy is not in configuration)
- resource "null_resource" "slack_hub_xindiy" {
-       id       = "1011018451728934545" -> null
-       triggers = {
          - "channel" = "C096TM8EMS8"
          - "user"    = "U0C1AAR21L5"
        } -> null
    }

Plan: 2 to add, 0 to change, 10 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 10 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.21s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_andyj will be created
+ resource "google_cloud_identity_group_membership" "tech_andyj" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "andyj@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

  # google_cloud_identity_group_membership.tech_aszhou will be destroyed
  # (because google_cloud_identity_group_membership.tech_aszhou is not in configuration)
- resource "google_cloud_identity_group_membership" "tech_aszhou" {
-       create_time = "2026-09-19T21:08:48.946958Z" -> null
-       group       = "groups/04anzqyu1bqbv79" -> null
-       id          = "groups/04anzqyu1bqbv79/memberships/100575886719005530375" -> null
-       name        = "groups/04anzqyu1bqbv79/memberships/100575886719005530375" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-19T21:08:48.946958Z" -> null

-       preferred_member_key {
-           id = "aszhou@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.tech_drishtis will be destroyed
  # (because google_cloud_identity_group_membership.tech_drishtis is not in configuration)
- resource "google_cloud_identity_group_membership" "tech_drishtis" {
-       create_time = "2026-09-19T21:08:48.948471Z" -> null
-       group       = "groups/04anzqyu1bqbv79" -> null
-       id          = "groups/04anzqyu1bqbv79/memberships/115738041677596308175" -> null
-       name        = "groups/04anzqyu1bqbv79/memberships/115738041677596308175" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-19T21:08:48.948471Z" -> null

-       preferred_member_key {
-           id = "drishtis@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.tech_indigo0445 will be destroyed
  # (because google_cloud_identity_group_membership.tech_indigo0445 is not in configuration)
- resource "google_cloud_identity_group_membership" "tech_indigo0445" {
-       create_time = "2026-09-19T21:08:48.954487Z" -> null
-       group       = "groups/04anzqyu1bqbv79" -> null
-       id          = "groups/04anzqyu1bqbv79/memberships/100660819832913404095" -> null
-       name        = "groups/04anzqyu1bqbv79/memberships/100660819832913404095" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-19T21:08:48.954487Z" -> null

-       preferred_member_key {
-           id = "rjwu@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.tech_xindiy will be destroyed
  # (because google_cloud_identity_group_membership.tech_xindiy is not in configuration)
- resource "google_cloud_identity_group_membership" "tech_xindiy" {
-       create_time = "2026-09-19T20:58:55.153034Z" -> null
-       group       = "groups/04anzqyu1bqbv79" -> null
-       id          = "groups/04anzqyu1bqbv79/memberships/100687050859054942101" -> null
-       name        = "groups/04anzqyu1bqbv79/memberships/100687050859054942101" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-19T20:58:55.153034Z" -> null

-       preferred_member_key {
-           id = "xindiy@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

Plan: 1 to add, 0 to change, 4 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 4 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-housing_housing" {
        id       = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d"
!       members  = [
-           "xindiy",
            # (13 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-study_study will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-study_study" {
        id       = "scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a"
!       members  = [
+           "andyj",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops will be updated in-place
! resource "keycloak_group_memberships" "project_devops" {
        id       = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7"
!       members  = [
-           "aszhou",
-           "azzhou",
-           "rjwu",
            # (3 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_changelog will be updated in-place
! resource "keycloak_group_memberships" "project_devops_changelog" {
        id       = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77"
!       members  = [
-           "aszhou",
-           "azzhou",
-           "rjwu",
            # (3 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_documentation will be updated in-place
! resource "keycloak_group_memberships" "project_devops_documentation" {
        id       = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f"
!       members  = [
-           "aszhou",
-           "azzhou",
-           "rjwu",
            # (3 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_link-shortener will be updated in-place
! resource "keycloak_group_memberships" "project_devops_link-shortener" {
        id       = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935"
!       members  = [
-           "aszhou",
-           "azzhou",
-           "rjwu",
            # (3 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_tartan-vote will be updated in-place
! resource "keycloak_group_memberships" "project_tartan-vote" {
        id       = "scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c"
!       members  = [
-           "drishtis",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
-           "aszhou",
-           "drishtis",
-           "rjwu",
-           "xindiy",
+           "andyj",
            # (36 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 8 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 8 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.26s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)
- destroy

OpenTofu will perform the following actions:

  # discord_member_roles.acchs will be updated in-place
! resource "discord_member_roles" "acchs" {
        id        = "1353118209024331948:781930226778767380"
        # (2 unchanged attributes hidden)

-       role {
-           has_role = true -> null
-           role_id  = "1488787087598813215" -> null
        }

        # (2 unchanged blocks hidden)
    }

  # discord_member_roles.andyj will be created
+ resource "discord_member_roles" "andyj" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "658742038233350194"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1550659990266581074"
        }
    }

  # discord_member_roles.aszhou will be destroyed
  # (because discord_member_roles.aszhou is not in configuration)
- resource "discord_member_roles" "aszhou" {
-       id        = "1353118209024331948:572567442153406479" -> null
-       server_id = "1353118209024331948" -> null
-       user_id   = "572567442153406479" -> null

-       role {
-           has_role = true -> null
-           role_id  = "1488787087598813215" -> null
        }
-       role {
-           has_role = true -> null
-           role_id  = "1488787088118644848" -> null
        }
    }

  # discord_member_roles.drishtis will be destroyed
  # (because discord_member_roles.drishtis is not in configuration)
- resource "discord_member_roles" "drishtis" {
-       id        = "1353118209024331948:1259254391635050518" -> null
-       server_id = "1353118209024331948" -> null
-       user_id   = "1259254391635050518" -> null

-       role {
-           has_role = true -> null
-           role_id  = "1488787088118644848" -> null
        }
-       role {
-           has_role = true -> null
-           role_id  = "1508601984255524885" -> null
        }
    }

  # discord_member_roles.indigo0445 will be destroyed
  # (because discord_member_roles.indigo0445 is not in configuration)
- resource "discord_member_roles" "indigo0445" {
-       id        = "1353118209024331948:629857326232371210" -> null
-       server_id = "1353118209024331948" -> null
-       user_id   = "629857326232371210" -> null

-       role {
-           has_role = true -> null
-           role_id  = "1488787087598813215" -> null
        }
-       role {
-           has_role = true -> null
-           role_id  = "1488787088118644848" -> null
        }
    }

  # discord_member_roles.xindiy will be destroyed
  # (because discord_member_roles.xindiy is not in configuration)
- resource "discord_member_roles" "xindiy" {
-       id        = "1353118209024331948:712363878851280936" -> null
-       server_id = "1353118209024331948" -> null
-       user_id   = "712363878851280936" -> null

-       role {
-           has_role = true -> null
-           role_id  = "1488787088118644848" -> null
        }
-       role {
-           has_role = true -> null
-           role_id  = "1513653499957874871" -> null
        }
    }

Plan: 1 to add, 1 to change, 4 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 1 to change, 4 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.21s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # forgejo_team.tech_leads will be destroyed
  # (because forgejo_team.tech_leads is not in configuration)
- resource "forgejo_team" "tech_leads" {
-       can_create_org_repo       = false -> null
-       description               = "All tech leads (write access to governance)" -> null
-       id                        = 44 -> null
-       includes_all_repositories = false -> null
-       name                      = "tech-leads" -> null
-       organization              = "ScottyLabs" -> null
-       organization_id           = 3 -> null
-       permission                = "write" -> null
-       units_map                 = {
          - "repo.code"   = "write"
          - "repo.issues" = "write"
          - "repo.pulls"  = "write"
        } -> null
    }

  # forgejo_team_member.cmu-housing_xindiy will be destroyed
  # (because forgejo_team_member.cmu-housing_xindiy is not in configuration)
- resource "forgejo_team_member" "cmu-housing_xindiy" {
-       team_id = 2 -> null
-       user    = "xindiy" -> null
    }

  # forgejo_team_member.cmu-study_andyj will be created
+ resource "forgejo_team_member" "cmu-study_andyj" {
+       team_id = 43
+       user    = "andyj"
    }

  # forgejo_team_member.devops_acchs will be destroyed
  # (because forgejo_team_member.devops_acchs is not in configuration)
- resource "forgejo_team_member" "devops_acchs" {
-       team_id = 4 -> null
-       user    = "acchs" -> null
    }

  # forgejo_team_member.devops_aszhou will be destroyed
  # (because forgejo_team_member.devops_aszhou is not in configuration)
- resource "forgejo_team_member" "devops_aszhou" {
-       team_id = 4 -> null
-       user    = "aszhou" -> null
    }

  # forgejo_team_member.devops_indigo0445 will be destroyed
  # (because forgejo_team_member.devops_indigo0445 is not in configuration)
- resource "forgejo_team_member" "devops_indigo0445" {
-       team_id = 4 -> null
-       user    = "rjwu" -> null
    }

  # forgejo_team_member.tartan-vote_drishtis will be destroyed
  # (because forgejo_team_member.tartan-vote_drishtis is not in configuration)
- resource "forgejo_team_member" "tartan-vote_drishtis" {
-       team_id = 6 -> null
-       user    = "drishtis" -> null
    }

  # forgejo_team_member.tech_leads_anish will be destroyed
  # (because forgejo_team_member.tech_leads_anish is not in configuration)
- resource "forgejo_team_member" "tech_leads_anish" {
-       team_id = 44 -> null
-       user    = "anish" -> null
    }

  # forgejo_team_member.tech_leads_annadavi will be destroyed
  # (because forgejo_team_member.tech_leads_annadavi is not in configuration)
- resource "forgejo_team_member" "tech_leads_annadavi" {
-       team_id = 44 -> null
-       user    = "annadavi" -> null
    }

  # forgejo_team_member.tech_leads_arom1a will be destroyed
  # (because forgejo_team_member.tech_leads_arom1a is not in configuration)
- resource "forgejo_team_member" "tech_leads_arom1a" {
-       team_id = 44 -> null
-       user    = "arom1a" -> null
    }

  # forgejo_team_member.tech_leads_dcrodrig will be destroyed
  # (because forgejo_team_member.tech_leads_dcrodrig is not in configuration)
- resource "forgejo_team_member" "tech_leads_dcrodrig" {
-       team_id = 44 -> null
-       user    = "dcrodrig" -> null
    }

  # forgejo_team_member.tech_leads_emmawong will be destroyed
  # (because forgejo_team_member.tech_leads_emmawong is not in configuration)
- resource "forgejo_team_member" "tech_leads_emmawong" {
-       team_id = 44 -> null
-       user    = "emmawong" -> null
    }

  # forgejo_team_member.tech_leads_etashjha will be destroyed
  # (because forgejo_team_member.tech_leads_etashjha is not in configuration)
- resource "forgejo_team_member" "tech_leads_etashjha" {
-       team_id = 44 -> null
-       user    = "etashjha" -> null
    }

  # forgejo_team_member.tech_leads_gostmeaper will be destroyed
  # (because forgejo_team_member.tech_leads_gostmeaper is not in configuration)
- resource "forgejo_team_member" "tech_leads_gostmeaper" {
-       team_id = 44 -> null
-       user    = "johncao" -> null
    }

  # forgejo_team_member.tech_leads_jalenluorion will be destroyed
  # (because forgejo_team_member.tech_leads_jalenluorion is not in configuration)
- resource "forgejo_team_member" "tech_leads_jalenluorion" {
-       team_id = 44 -> null
-       user    = "jalenluorion" -> null
    }

  # forgejo_team_member.tech_leads_krishsax will be destroyed
  # (because forgejo_team_member.tech_leads_krishsax is not in configuration)
- resource "forgejo_team_member" "tech_leads_krishsax" {
-       team_id = 44 -> null
-       user    = "krishsax" -> null
    }

  # forgejo_team_member.tech_leads_kritdass will be destroyed
  # (because forgejo_team_member.tech_leads_kritdass is not in configuration)
- resource "forgejo_team_member" "tech_leads_kritdass" {
-       team_id = 44 -> null
-       user    = "kdass" -> null
    }

  # forgejo_team_member.tech_leads_maybe_yiyi will be destroyed
  # (because forgejo_team_member.tech_leads_maybe_yiyi is not in configuration)
- resource "forgejo_team_member" "tech_leads_maybe_yiyi" {
-       team_id = 44 -> null
-       user    = "yiyoungl" -> null
    }

  # forgejo_team_member.tech_leads_nikhilr will be destroyed
  # (because forgejo_team_member.tech_leads_nikhilr is not in configuration)
- resource "forgejo_team_member" "tech_leads_nikhilr" {
-       team_id = 44 -> null
-       user    = "nramanuj" -> null
    }

  # forgejo_team_member.tech_leads_siddharthr will be destroyed
  # (because forgejo_team_member.tech_leads_siddharthr is not in configuration)
- resource "forgejo_team_member" "tech_leads_siddharthr" {
-       team_id = 44 -> null
-       user    = "sradhak2" -> null
    }

  # forgejo_team_member.tech_leads_tentype will be destroyed
  # (because forgejo_team_member.tech_leads_tentype is not in configuration)
- resource "forgejo_team_member" "tech_leads_tentype" {
-       team_id = 44 -> null
-       user    = "maxwen" -> null
    }

  # forgejo_team_member.tech_leads_thesuperrl will be destroyed
  # (because forgejo_team_member.tech_leads_thesuperrl is not in configuration)
- resource "forgejo_team_member" "tech_leads_thesuperrl" {
-       team_id = 44 -> null
-       user    = "thesuperrl" -> null
    }

  # forgejo_team_member.tech_leads_xboxbedrock will be destroyed
  # (because forgejo_team_member.tech_leads_xboxbedrock is not in configuration)
- resource "forgejo_team_member" "tech_leads_xboxbedrock" {
-       team_id = 44 -> null
-       user    = "xboxbedrock" -> null
    }

  # forgejo_team_repository.tech_leads_governance will be destroyed
  # (because forgejo_team_repository.tech_leads_governance is not in configuration)
- resource "forgejo_team_repository" "tech_leads_governance" {
-       owner      = "ScottyLabs" -> null
-       repository = "governance" -> null
-       team_id    = 44 -> null
    }

  # github_team_membership.cmu-housing_xindiy will be destroyed
  # (because github_team_membership.cmu-housing_xindiy is not in configuration)
- resource "github_team_membership" "cmu-housing_xindiy" {
-       etag     = "W/\"4f9220513935c71512fd7ec92aedb76b04ceb29a03ad5312213f818db1cd729e\"" -> null
-       id       = "17937919:xindiyang0" -> null
-       role     = "member" -> null
-       team_id  = "17937919" -> null
-       username = "xindiyang0" -> null
    }

  # github_team_membership.cmu-study_andyj will be created
+ resource "github_team_membership" "cmu-study_andyj" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "19579842"
+       username = "anan193"
    }

  # github_team_membership.devops_acchs will be destroyed
  # (because github_team_membership.devops_acchs is not in configuration)
- resource "github_team_membership" "devops_acchs" {
-       etag     = "W/\"4179f054de82f27572bb3c7ea816b563cafc9fbc4d35a96fb398e60b043d44d1\"" -> null
-       id       = "14706218:acchs-at-cmu" -> null
-       role     = "member" -> null
-       team_id  = "14706218" -> null
-       username = "acchs-at-cmu" -> null
    }

  # github_team_membership.devops_aszhou will be destroyed
  # (because github_team_membership.devops_aszhou is not in configuration)
- resource "github_team_membership" "devops_aszhou" {
-       etag     = "W/\"7ded4c758e7d683f26bbd6ba5249b0169eb642c18bd932c07b551232053c34d8\"" -> null
-       id       = "14706218:azhou22" -> null
-       role     = "member" -> null
-       team_id  = "14706218" -> null
-       username = "azhou22" -> null
    }

  # github_team_membership.devops_indigo0445 will be destroyed
  # (because github_team_membership.devops_indigo0445 is not in configuration)
- resource "github_team_membership" "devops_indigo0445" {
-       etag     = "W/\"b583553f0de8ce760b6183f9442faaf221b0085c93ef475142722a4d1f324c8b\"" -> null
-       id       = "14706218:indigo0445" -> null
-       role     = "member" -> null
-       team_id  = "14706218" -> null
-       username = "indigo0445" -> null
    }

  # github_team_membership.tartan-vote_drishtis will be destroyed
  # (because github_team_membership.tartan-vote_drishtis is not in configuration)
- resource "github_team_membership" "tartan-vote_drishtis" {
-       etag     = "W/\"61d7dcdff5a005a02cae3352bc5b41d7e5f8142e002b279558301743382729f6\"" -> null
-       id       = "17701833:drishtisingh489" -> null
-       role     = "member" -> null
-       team_id  = "17701833" -> null
-       username = "drishtisingh489" -> null
    }

Plan: 2 to add, 0 to change, 28 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 0 to change, 28 to destroy.


Plan Summary

6 projects, 5 with changes, 1 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.29s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # null_resource.slack_cmu-study_ch0_andyj will be created + resource "null_resource" "slack_cmu-study_ch0_andyj" { + id = (known after apply) + triggers = { + "channel" = "C09KN6M3P53" + "user" = "U0C2066T7T9" } } # null_resource.slack_devops_ch0_acchs will be destroyed # (because null_resource.slack_devops_ch0_acchs is not in configuration) - resource "null_resource" "slack_devops_ch0_acchs" { - id = "5494655407757332457" -> null - triggers = { - "channel" = "C08K3Q77ZQF" - "user" = "U0BVBNX8YGZ" } -> null } # null_resource.slack_devops_ch0_aszhou will be destroyed # (because null_resource.slack_devops_ch0_aszhou is not in configuration) - resource "null_resource" "slack_devops_ch0_aszhou" { - id = "3525995290209722166" -> null - triggers = { - "channel" = "C08K3Q77ZQF" - "user" = "U09CQ04V4BT" } -> null } # null_resource.slack_devops_ch0_indigo0445 will be destroyed # (because null_resource.slack_devops_ch0_indigo0445 is not in configuration) - resource "null_resource" "slack_devops_ch0_indigo0445" { - id = "166910966562392614" -> null - triggers = { - "channel" = "C08K3Q77ZQF" - "user" = "U09C4RTR36F" } -> null } # null_resource.slack_devops_ch1_acchs will be destroyed # (because null_resource.slack_devops_ch1_acchs is not in configuration) - resource "null_resource" "slack_devops_ch1_acchs" { - id = "1105780819490615240" -> null - triggers = { - "channel" = "C0BNV84NUTC" - "user" = "U0BVBNX8YGZ" } -> null } # null_resource.slack_devops_ch1_aszhou will be destroyed # (because null_resource.slack_devops_ch1_aszhou is not in configuration) - resource "null_resource" "slack_devops_ch1_aszhou" { - id = "4874526058879346740" -> null - triggers = { - "channel" = "C0BNV84NUTC" - "user" = "U09CQ04V4BT" } -> null } # null_resource.slack_devops_ch1_indigo0445 will be destroyed # (because null_resource.slack_devops_ch1_indigo0445 is not in configuration) - resource "null_resource" "slack_devops_ch1_indigo0445" { - id = "7985592316574563315" -> null - triggers = { - "channel" = "C0BNV84NUTC" - "user" = "U09C4RTR36F" } -> null } # null_resource.slack_hub_andyj will be created + resource "null_resource" "slack_hub_andyj" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0C2066T7T9" } } # null_resource.slack_hub_aszhou will be destroyed # (because null_resource.slack_hub_aszhou is not in configuration) - resource "null_resource" "slack_hub_aszhou" { - id = "2233394787102757076" -> null - triggers = { - "channel" = "C096TM8EMS8" - "user" = "U09CQ04V4BT" } -> null } # null_resource.slack_hub_drishtis will be destroyed # (because null_resource.slack_hub_drishtis is not in configuration) - resource "null_resource" "slack_hub_drishtis" { - id = "465737434924418541" -> null - triggers = { - "channel" = "C096TM8EMS8" - "user" = "U0BTH04N9SP" } -> null } # null_resource.slack_hub_indigo0445 will be destroyed # (because null_resource.slack_hub_indigo0445 is not in configuration) - resource "null_resource" "slack_hub_indigo0445" { - id = "5455653694878708973" -> null - triggers = { - "channel" = "C096TM8EMS8" - "user" = "U09C4RTR36F" } -> null } # null_resource.slack_hub_xindiy will be destroyed # (because null_resource.slack_hub_xindiy is not in configuration) - resource "null_resource" "slack_hub_xindiy" { - id = "1011018451728934545" -> null - triggers = { - "channel" = "C096TM8EMS8" - "user" = "U0C1AAR21L5" } -> null } Plan: 2 to add, 0 to change, 10 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 10 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.21s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_andyj will be created + resource "google_cloud_identity_group_membership" "tech_andyj" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "andyj@andrew.cmu.edu" } + roles { + name = "MEMBER" } } # google_cloud_identity_group_membership.tech_aszhou will be destroyed # (because google_cloud_identity_group_membership.tech_aszhou is not in configuration) - resource "google_cloud_identity_group_membership" "tech_aszhou" { - create_time = "2026-09-19T21:08:48.946958Z" -> null - group = "groups/04anzqyu1bqbv79" -> null - id = "groups/04anzqyu1bqbv79/memberships/100575886719005530375" -> null - name = "groups/04anzqyu1bqbv79/memberships/100575886719005530375" -> null - type = "USER" -> null - update_time = "2026-09-19T21:08:48.946958Z" -> null - preferred_member_key { - id = "aszhou@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.tech_drishtis will be destroyed # (because google_cloud_identity_group_membership.tech_drishtis is not in configuration) - resource "google_cloud_identity_group_membership" "tech_drishtis" { - create_time = "2026-09-19T21:08:48.948471Z" -> null - group = "groups/04anzqyu1bqbv79" -> null - id = "groups/04anzqyu1bqbv79/memberships/115738041677596308175" -> null - name = "groups/04anzqyu1bqbv79/memberships/115738041677596308175" -> null - type = "USER" -> null - update_time = "2026-09-19T21:08:48.948471Z" -> null - preferred_member_key { - id = "drishtis@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.tech_indigo0445 will be destroyed # (because google_cloud_identity_group_membership.tech_indigo0445 is not in configuration) - resource "google_cloud_identity_group_membership" "tech_indigo0445" { - create_time = "2026-09-19T21:08:48.954487Z" -> null - group = "groups/04anzqyu1bqbv79" -> null - id = "groups/04anzqyu1bqbv79/memberships/100660819832913404095" -> null - name = "groups/04anzqyu1bqbv79/memberships/100660819832913404095" -> null - type = "USER" -> null - update_time = "2026-09-19T21:08:48.954487Z" -> null - preferred_member_key { - id = "rjwu@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.tech_xindiy will be destroyed # (because google_cloud_identity_group_membership.tech_xindiy is not in configuration) - resource "google_cloud_identity_group_membership" "tech_xindiy" { - create_time = "2026-09-19T20:58:55.153034Z" -> null - group = "groups/04anzqyu1bqbv79" -> null - id = "groups/04anzqyu1bqbv79/memberships/100687050859054942101" -> null - name = "groups/04anzqyu1bqbv79/memberships/100687050859054942101" -> null - type = "USER" -> null - update_time = "2026-09-19T20:58:55.153034Z" -> null - preferred_member_key { - id = "xindiy@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } Plan: 1 to add, 0 to change, 4 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 4 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-housing_housing" { id = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d" ! members = [ - "xindiy", # (13 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-study_study will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-study_study" { id = "scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a" ! members = [ + "andyj", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops will be updated in-place ! resource "keycloak_group_memberships" "project_devops" { id = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7" ! members = [ - "aszhou", - "azzhou", - "rjwu", # (3 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_changelog will be updated in-place ! resource "keycloak_group_memberships" "project_devops_changelog" { id = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77" ! members = [ - "aszhou", - "azzhou", - "rjwu", # (3 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_documentation will be updated in-place ! resource "keycloak_group_memberships" "project_devops_documentation" { id = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f" ! members = [ - "aszhou", - "azzhou", - "rjwu", # (3 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_link-shortener will be updated in-place ! resource "keycloak_group_memberships" "project_devops_link-shortener" { id = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935" ! members = [ - "aszhou", - "azzhou", - "rjwu", # (3 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_tartan-vote will be updated in-place ! resource "keycloak_group_memberships" "project_tartan-vote" { id = "scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c" ! members = [ - "drishtis", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ - "aszhou", - "drishtis", - "rjwu", - "xindiy", + "andyj", # (36 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 8 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 8 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.26s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) - destroy OpenTofu will perform the following actions: # discord_member_roles.acchs will be updated in-place ! resource "discord_member_roles" "acchs" { id = "1353118209024331948:781930226778767380" # (2 unchanged attributes hidden) - role { - has_role = true -> null - role_id = "1488787087598813215" -> null } # (2 unchanged blocks hidden) } # discord_member_roles.andyj will be created + resource "discord_member_roles" "andyj" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "658742038233350194" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1550659990266581074" } } # discord_member_roles.aszhou will be destroyed # (because discord_member_roles.aszhou is not in configuration) - resource "discord_member_roles" "aszhou" { - id = "1353118209024331948:572567442153406479" -> null - server_id = "1353118209024331948" -> null - user_id = "572567442153406479" -> null - role { - has_role = true -> null - role_id = "1488787087598813215" -> null } - role { - has_role = true -> null - role_id = "1488787088118644848" -> null } } # discord_member_roles.drishtis will be destroyed # (because discord_member_roles.drishtis is not in configuration) - resource "discord_member_roles" "drishtis" { - id = "1353118209024331948:1259254391635050518" -> null - server_id = "1353118209024331948" -> null - user_id = "1259254391635050518" -> null - role { - has_role = true -> null - role_id = "1488787088118644848" -> null } - role { - has_role = true -> null - role_id = "1508601984255524885" -> null } } # discord_member_roles.indigo0445 will be destroyed # (because discord_member_roles.indigo0445 is not in configuration) - resource "discord_member_roles" "indigo0445" { - id = "1353118209024331948:629857326232371210" -> null - server_id = "1353118209024331948" -> null - user_id = "629857326232371210" -> null - role { - has_role = true -> null - role_id = "1488787087598813215" -> null } - role { - has_role = true -> null - role_id = "1488787088118644848" -> null } } # discord_member_roles.xindiy will be destroyed # (because discord_member_roles.xindiy is not in configuration) - resource "discord_member_roles" "xindiy" { - id = "1353118209024331948:712363878851280936" -> null - server_id = "1353118209024331948" -> null - user_id = "712363878851280936" -> null - role { - has_role = true -> null - role_id = "1488787088118644848" -> null } - role { - has_role = true -> null - role_id = "1513653499957874871" -> null } } Plan: 1 to add, 1 to change, 4 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 1 to change, 4 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.21s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # forgejo_team.tech_leads will be destroyed # (because forgejo_team.tech_leads is not in configuration) - resource "forgejo_team" "tech_leads" { - can_create_org_repo = false -> null - description = "All tech leads (write access to governance)" -> null - id = 44 -> null - includes_all_repositories = false -> null - name = "tech-leads" -> null - organization = "ScottyLabs" -> null - organization_id = 3 -> null - permission = "write" -> null - units_map = { - "repo.code" = "write" - "repo.issues" = "write" - "repo.pulls" = "write" } -> null } # forgejo_team_member.cmu-housing_xindiy will be destroyed # (because forgejo_team_member.cmu-housing_xindiy is not in configuration) - resource "forgejo_team_member" "cmu-housing_xindiy" { - team_id = 2 -> null - user = "xindiy" -> null } # forgejo_team_member.cmu-study_andyj will be created + resource "forgejo_team_member" "cmu-study_andyj" { + team_id = 43 + user = "andyj" } # forgejo_team_member.devops_acchs will be destroyed # (because forgejo_team_member.devops_acchs is not in configuration) - resource "forgejo_team_member" "devops_acchs" { - team_id = 4 -> null - user = "acchs" -> null } # forgejo_team_member.devops_aszhou will be destroyed # (because forgejo_team_member.devops_aszhou is not in configuration) - resource "forgejo_team_member" "devops_aszhou" { - team_id = 4 -> null - user = "aszhou" -> null } # forgejo_team_member.devops_indigo0445 will be destroyed # (because forgejo_team_member.devops_indigo0445 is not in configuration) - resource "forgejo_team_member" "devops_indigo0445" { - team_id = 4 -> null - user = "rjwu" -> null } # forgejo_team_member.tartan-vote_drishtis will be destroyed # (because forgejo_team_member.tartan-vote_drishtis is not in configuration) - resource "forgejo_team_member" "tartan-vote_drishtis" { - team_id = 6 -> null - user = "drishtis" -> null } # forgejo_team_member.tech_leads_anish will be destroyed # (because forgejo_team_member.tech_leads_anish is not in configuration) - resource "forgejo_team_member" "tech_leads_anish" { - team_id = 44 -> null - user = "anish" -> null } # forgejo_team_member.tech_leads_annadavi will be destroyed # (because forgejo_team_member.tech_leads_annadavi is not in configuration) - resource "forgejo_team_member" "tech_leads_annadavi" { - team_id = 44 -> null - user = "annadavi" -> null } # forgejo_team_member.tech_leads_arom1a will be destroyed # (because forgejo_team_member.tech_leads_arom1a is not in configuration) - resource "forgejo_team_member" "tech_leads_arom1a" { - team_id = 44 -> null - user = "arom1a" -> null } # forgejo_team_member.tech_leads_dcrodrig will be destroyed # (because forgejo_team_member.tech_leads_dcrodrig is not in configuration) - resource "forgejo_team_member" "tech_leads_dcrodrig" { - team_id = 44 -> null - user = "dcrodrig" -> null } # forgejo_team_member.tech_leads_emmawong will be destroyed # (because forgejo_team_member.tech_leads_emmawong is not in configuration) - resource "forgejo_team_member" "tech_leads_emmawong" { - team_id = 44 -> null - user = "emmawong" -> null } # forgejo_team_member.tech_leads_etashjha will be destroyed # (because forgejo_team_member.tech_leads_etashjha is not in configuration) - resource "forgejo_team_member" "tech_leads_etashjha" { - team_id = 44 -> null - user = "etashjha" -> null } # forgejo_team_member.tech_leads_gostmeaper will be destroyed # (because forgejo_team_member.tech_leads_gostmeaper is not in configuration) - resource "forgejo_team_member" "tech_leads_gostmeaper" { - team_id = 44 -> null - user = "johncao" -> null } # forgejo_team_member.tech_leads_jalenluorion will be destroyed # (because forgejo_team_member.tech_leads_jalenluorion is not in configuration) - resource "forgejo_team_member" "tech_leads_jalenluorion" { - team_id = 44 -> null - user = "jalenluorion" -> null } # forgejo_team_member.tech_leads_krishsax will be destroyed # (because forgejo_team_member.tech_leads_krishsax is not in configuration) - resource "forgejo_team_member" "tech_leads_krishsax" { - team_id = 44 -> null - user = "krishsax" -> null } # forgejo_team_member.tech_leads_kritdass will be destroyed # (because forgejo_team_member.tech_leads_kritdass is not in configuration) - resource "forgejo_team_member" "tech_leads_kritdass" { - team_id = 44 -> null - user = "kdass" -> null } # forgejo_team_member.tech_leads_maybe_yiyi will be destroyed # (because forgejo_team_member.tech_leads_maybe_yiyi is not in configuration) - resource "forgejo_team_member" "tech_leads_maybe_yiyi" { - team_id = 44 -> null - user = "yiyoungl" -> null } # forgejo_team_member.tech_leads_nikhilr will be destroyed # (because forgejo_team_member.tech_leads_nikhilr is not in configuration) - resource "forgejo_team_member" "tech_leads_nikhilr" { - team_id = 44 -> null - user = "nramanuj" -> null } # forgejo_team_member.tech_leads_siddharthr will be destroyed # (because forgejo_team_member.tech_leads_siddharthr is not in configuration) - resource "forgejo_team_member" "tech_leads_siddharthr" { - team_id = 44 -> null - user = "sradhak2" -> null } # forgejo_team_member.tech_leads_tentype will be destroyed # (because forgejo_team_member.tech_leads_tentype is not in configuration) - resource "forgejo_team_member" "tech_leads_tentype" { - team_id = 44 -> null - user = "maxwen" -> null } # forgejo_team_member.tech_leads_thesuperrl will be destroyed # (because forgejo_team_member.tech_leads_thesuperrl is not in configuration) - resource "forgejo_team_member" "tech_leads_thesuperrl" { - team_id = 44 -> null - user = "thesuperrl" -> null } # forgejo_team_member.tech_leads_xboxbedrock will be destroyed # (because forgejo_team_member.tech_leads_xboxbedrock is not in configuration) - resource "forgejo_team_member" "tech_leads_xboxbedrock" { - team_id = 44 -> null - user = "xboxbedrock" -> null } # forgejo_team_repository.tech_leads_governance will be destroyed # (because forgejo_team_repository.tech_leads_governance is not in configuration) - resource "forgejo_team_repository" "tech_leads_governance" { - owner = "ScottyLabs" -> null - repository = "governance" -> null - team_id = 44 -> null } # github_team_membership.cmu-housing_xindiy will be destroyed # (because github_team_membership.cmu-housing_xindiy is not in configuration) - resource "github_team_membership" "cmu-housing_xindiy" { - etag = "W/\"4f9220513935c71512fd7ec92aedb76b04ceb29a03ad5312213f818db1cd729e\"" -> null - id = "17937919:xindiyang0" -> null - role = "member" -> null - team_id = "17937919" -> null - username = "xindiyang0" -> null } # github_team_membership.cmu-study_andyj will be created + resource "github_team_membership" "cmu-study_andyj" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "19579842" + username = "anan193" } # github_team_membership.devops_acchs will be destroyed # (because github_team_membership.devops_acchs is not in configuration) - resource "github_team_membership" "devops_acchs" { - etag = "W/\"4179f054de82f27572bb3c7ea816b563cafc9fbc4d35a96fb398e60b043d44d1\"" -> null - id = "14706218:acchs-at-cmu" -> null - role = "member" -> null - team_id = "14706218" -> null - username = "acchs-at-cmu" -> null } # github_team_membership.devops_aszhou will be destroyed # (because github_team_membership.devops_aszhou is not in configuration) - resource "github_team_membership" "devops_aszhou" { - etag = "W/\"7ded4c758e7d683f26bbd6ba5249b0169eb642c18bd932c07b551232053c34d8\"" -> null - id = "14706218:azhou22" -> null - role = "member" -> null - team_id = "14706218" -> null - username = "azhou22" -> null } # github_team_membership.devops_indigo0445 will be destroyed # (because github_team_membership.devops_indigo0445 is not in configuration) - resource "github_team_membership" "devops_indigo0445" { - etag = "W/\"b583553f0de8ce760b6183f9442faaf221b0085c93ef475142722a4d1f324c8b\"" -> null - id = "14706218:indigo0445" -> null - role = "member" -> null - team_id = "14706218" -> null - username = "indigo0445" -> null } # github_team_membership.tartan-vote_drishtis will be destroyed # (because github_team_membership.tartan-vote_drishtis is not in configuration) - resource "github_team_membership" "tartan-vote_drishtis" { - etag = "W/\"61d7dcdff5a005a02cae3352bc5b41d7e5f8142e002b279558301743382729f6\"" -> null - id = "17701833:drishtisingh489" -> null - role = "member" -> null - team_id = "17701833" -> null - username = "drishtisingh489" -> null } Plan: 2 to add, 0 to change, 28 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 0 to change, 28 to destroy. --- ### Plan Summary 6 projects, 5 with changes, 1 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
xboxbedrock force-pushed add-name from b07dd3f370
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: slack Plan: 2 to add, 0 to change, 10 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 4 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 8 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 1 to change, 4 to destroy.
atlantis/plan: git Plan: 2 to add, 0 to change, 28 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
to 4c6f89ac4d
All checks were successful
atlantis/plan: slack Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 2 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
atlantis/plan: git Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
2026-09-19 21:10:35 +00:00
Compare

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.71s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-study_ch0_andyj will be created
+ resource "null_resource" "slack_cmu-study_ch0_andyj" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C09KN6M3P53"
          + "user"    = "U0C2066T7T9"
        }
    }

  # null_resource.slack_hub_andyj will be created
+ resource "null_resource" "slack_hub_andyj" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0C2066T7T9"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 0 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.75s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.69s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_andyj will be created
+ resource "google_cloud_identity_group_membership" "tech_andyj" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "andyj@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/225/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/225/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/225/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 7.68s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-study_study will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-study_study" {
        id       = "scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a"
!       members  = [
+           "andyj",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "andyj",
            # (40 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 2 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 2 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.72s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.andyj will be created
+ resource "discord_member_roles" "andyj" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "658742038233350194"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1550659990266581074"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.67s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # forgejo_team_member.cmu-study_andyj will be created
+ resource "forgejo_team_member" "cmu-study_andyj" {
+       team_id = 43
+       user    = "andyj"
    }

  # github_team_membership.cmu-study_andyj will be created
+ resource "github_team_membership" "cmu-study_andyj" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "19579842"
+       username = "anan193"
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 0 to change, 0 to destroy.


Plan Summary

6 projects, 5 with changes, 1 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.71s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_cmu-study_ch0_andyj will be created + resource "null_resource" "slack_cmu-study_ch0_andyj" { + id = (known after apply) + triggers = { + "channel" = "C09KN6M3P53" + "user" = "U0C2066T7T9" } } # null_resource.slack_hub_andyj will be created + resource "null_resource" "slack_hub_andyj" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0C2066T7T9" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.75s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.69s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_andyj will be created + resource "google_cloud_identity_group_membership" "tech_andyj" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "andyj@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/225/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/225/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/225/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 7.68s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-study_study will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-study_study" { id = "scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a" ! members = [ + "andyj", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "andyj", # (40 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 2 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 2 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.72s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.andyj will be created + resource "discord_member_roles" "andyj" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "658742038233350194" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1550659990266581074" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.67s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # forgejo_team_member.cmu-study_andyj will be created + resource "forgejo_team_member" "cmu-study_andyj" { + team_id = 43 + user = "andyj" } # github_team_membership.cmu-study_andyj will be created + resource "github_team_membership" "cmu-study_andyj" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "19579842" + username = "anan193" } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### Plan Summary 6 projects, 5 with changes, 1 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis apply

atlantis apply

Ran Apply for 6 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: keycloak dir: tofu/keycloak workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.17s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

2. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.16s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_cmu-study_ch0_andyj: Creating...
null_resource.slack_hub_andyj: Creating...
null_resource.slack_hub_andyj: Provisioning with 'local-exec'...
null_resource.slack_hub_andyj (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C2066T7T9"]
null_resource.slack_cmu-study_ch0_andyj: Provisioning with 'local-exec'...
null_resource.slack_cmu-study_ch0_andyj (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C09KN6M3P53 --user U0C2066T7T9"]
null_resource.slack_cmu-study_ch0_andyj: Creation complete after 0s [id=3222493524866225663]
null_resource.slack_hub_andyj: Creation complete after 0s [id=1529657801330481441]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

3. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.13s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.andyj: Creating...
discord_member_roles.andyj: Creation complete after 1s [id=1353118209024331948:658742038233350194]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.16s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_andyj: Creating...
google_cloud_identity_group_membership.tech_andyj: Creation complete after 0s [id=groups/04anzqyu1bqbv79/memberships/103852186490384222635]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.22s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
github_team_membership.cmu-study_andyj: Creating...
forgejo_team_member.cmu-study_andyj: Creating...
forgejo_team_member.cmu-study_andyj: Creation complete after 0s
github_team_membership.cmu-study_andyj: Creation complete after 2s [id=19579842:anan193]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

6. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cmu-study_study: Modifying... [id=scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_cmu-study_study: Modifications complete after 2s [id=scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a]
keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 2 changed, 0 destroyed.

Apply Summary

6 projects, 6 successful, 0 failed, 0 errored

Ran Apply for 6 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.17s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.16s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_cmu-study_ch0_andyj: Creating... null_resource.slack_hub_andyj: Creating... null_resource.slack_hub_andyj: Provisioning with 'local-exec'... null_resource.slack_hub_andyj (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C2066T7T9"] null_resource.slack_cmu-study_ch0_andyj: Provisioning with 'local-exec'... null_resource.slack_cmu-study_ch0_andyj (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C09KN6M3P53 --user U0C2066T7T9"] null_resource.slack_cmu-study_ch0_andyj: Creation complete after 0s [id=3222493524866225663] null_resource.slack_hub_andyj: Creation complete after 0s [id=1529657801330481441] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.13s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.andyj: Creating... discord_member_roles.andyj: Creation complete after 1s [id=1353118209024331948:658742038233350194] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.16s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_andyj: Creating... google_cloud_identity_group_membership.tech_andyj: Creation complete after 0s [id=groups/04anzqyu1bqbv79/memberships/103852186490384222635] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.22s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ github_team_membership.cmu-study_andyj: Creating... forgejo_team_member.cmu-study_andyj: Creating... forgejo_team_member.cmu-study_andyj: Creation complete after 0s github_team_membership.cmu-study_andyj: Creation complete after 2s [id=19579842:anan193] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 6. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cmu-study_study: Modifying... [id=scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_cmu-study_study: Modifications complete after 2s [id=scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a] keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 2 changed, 0 destroyed. ``` </details> --- ### Apply Summary 6 projects, 6 successful, 0 failed, 0 errored

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
brightz referenced this pull request from a commit 2026-09-26 22:03:28 +00:00
Sign in to join this conversation.
No description provided.