fix: let tech leads have real write access for merging #232
No reviewers
Labels
No labels
needs/linking
needs/docs
needs/research
needs/upstream
type/bug
type/external
type/feature
type/refactor
urgency
high
urgency
immediate
urgency
low
urgency
medium
No milestone
No project
No assignees
2 participants
Notifications
Due date
No due date set.
Dependencies
No dependencies set
Reference
ScottyLabs/governance!232
Loading…
Reference in a new issue
No description provided.
Delete branch "forgejo-tech-leads-write"
Deleting a branch is permanent. Although the deleted branch may continue to exist for a short time before it actually gets removed, it CANNOT be undone in most cases. Continue?
Ran Plan for dir:
tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.atlantis plan
Ran Plan for dir:
tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.atlantis plan
Ran Plan for dir:
tofu/gitworkspace:defaultPlan Error
Show Output
atlantis plan
Ran Plan for dir:
tofu/gitworkspace:defaultPlan Error
Show Output
8085733afbbffea6fdf4Ran Plan for dir:
tofu/gitworkspace:defaultPlan Error
Show Output
atlantis plan
Ran Plan for dir:
tofu/gitworkspace:defaultPlan Error
Show Output
bffea6fdf48d4a2cdda6Ran Plan for project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 18 to add, 0 to change, 0 to destroy.
atlantis apply
Ran Apply for project:
gitdir:tofu/gitworkspace:defaultShow Output
Locks and plans deleted for the projects and workspaces modified in this pull request:
tofu/gitworkspace:default