feat: add indigo0445 to devops #226

Merged
xboxbedrock merged 3 commits from indigo0445/governance:add-indigo0445-member into main 2026-09-19 20:51:27 +00:00
Member

new

new
add member indigo0445 to devops
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 2m51s
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
CI / build (pull_request) Has been cancelled
527a26b45e
indigo0445 changed title from add member indigo0445 to devops to feat: add member indigo0445 to devops 2026-09-19 04:37:57 +00:00
indigo0445 changed title from feat: add member indigo0445 to devops to feat: add indigo0445 to devops 2026-09-19 04:38:24 +00:00
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: slack Plan: 3 to add, 0 to change, 5 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 1 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 5 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 1 to change, 1 to destroy.
atlantis/plan: git Plan: 2 to add, 1 to change, 22 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
bbc2bd3525

Ran Plan for 6 projects:

  1. project: google_groups dir: tofu/google_groups workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: git dir: tofu/git workspace: default
  4. project: discord dir: tofu/discord workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default
  6. project: google_play dir: tofu/google_play workspace: default

1. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` --- ### 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #217. To continue, delete the lock from #217 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: google_groups dir: tofu/google_groups workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: google_play dir: tofu/google_play workspace: default
  4. project: git dir: tofu/git workspace: default
  5. project: slack dir: tofu/slack workspace: default
  6. project: keycloak dir: tofu/keycloak workspace: default

1. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.34s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_devops_ch0_acchs will be destroyed
  # (because null_resource.slack_devops_ch0_acchs is not in configuration)
- resource "null_resource" "slack_devops_ch0_acchs" {
-       id       = "5494655407757332457" -> null
-       triggers = {
          - "channel" = "C08K3Q77ZQF"
          - "user"    = "U0BVBNX8YGZ"
        } -> null
    }

  # null_resource.slack_devops_ch0_indigo0445 will be created
+ resource "null_resource" "slack_devops_ch0_indigo0445" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08K3Q77ZQF"
          + "user"    = "U09C4RTR36F"
        }
    }

  # null_resource.slack_devops_ch1_acchs will be destroyed
  # (because null_resource.slack_devops_ch1_acchs is not in configuration)
- resource "null_resource" "slack_devops_ch1_acchs" {
-       id       = "1105780819490615240" -> null
-       triggers = {
          - "channel" = "C0BNV84NUTC"
          - "user"    = "U0BVBNX8YGZ"
        } -> null
    }

  # null_resource.slack_devops_ch1_indigo0445 will be created
+ resource "null_resource" "slack_devops_ch1_indigo0445" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0BNV84NUTC"
          + "user"    = "U09C4RTR36F"
        }
    }

  # null_resource.slack_hub_indigo0445 will be created
+ resource "null_resource" "slack_hub_indigo0445" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U09C4RTR36F"
        }
    }

Plan: 3 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 3 to add, 0 to change, 2 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.29s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on shared package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.39s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_indigo0445 will be created
+ resource "google_cloud_identity_group_membership" "tech_indigo0445" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "rjwu@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.31s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_devops will be updated in-place
! resource "keycloak_group_memberships" "project_devops" {
        id       = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7"
!       members  = [
-           "azzhou",
+           "rjwu",
            # (3 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_changelog will be updated in-place
! resource "keycloak_group_memberships" "project_devops_changelog" {
        id       = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77"
!       members  = [
-           "azzhou",
+           "rjwu",
            # (3 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_documentation will be updated in-place
! resource "keycloak_group_memberships" "project_devops_documentation" {
        id       = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f"
!       members  = [
-           "azzhou",
+           "rjwu",
            # (3 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_link-shortener will be updated in-place
! resource "keycloak_group_memberships" "project_devops_link-shortener" {
        id       = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935"
!       members  = [
-           "azzhou",
+           "rjwu",
            # (3 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "rjwu",
            # (36 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 5 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 5 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.32s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # discord_member_roles.acchs will be updated in-place
! resource "discord_member_roles" "acchs" {
        id        = "1353118209024331948:781930226778767380"
        # (2 unchanged attributes hidden)

-       role {
-           has_role = true -> null
-           role_id  = "1488787087598813215" -> null
        }

        # (2 unchanged blocks hidden)
    }

  # discord_member_roles.indigo0445 will be created
+ resource "discord_member_roles" "indigo0445" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "629857326232371210"

+       role {
+           has_role = true
+           role_id  = "1488787087598813215"
        }
+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
    }

Plan: 1 to add, 1 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 1 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.39s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)
- destroy

OpenTofu will perform the following actions:

  # forgejo_team.tech_leads will be destroyed
  # (because forgejo_team.tech_leads is not in configuration)
- resource "forgejo_team" "tech_leads" {
-       can_create_org_repo       = false -> null
-       description               = "All tech leads (write access to governance)" -> null
-       id                        = 44 -> null
-       includes_all_repositories = false -> null
-       name                      = "tech-leads" -> null
-       organization              = "ScottyLabs" -> null
-       organization_id           = 3 -> null
-       permission                = "write" -> null
-       units_map                 = {
          - "repo.code"   = "write"
          - "repo.issues" = "write"
          - "repo.pulls"  = "write"
        } -> null
    }

  # forgejo_team_member.devops_acchs will be destroyed
  # (because forgejo_team_member.devops_acchs is not in configuration)
- resource "forgejo_team_member" "devops_acchs" {
-       team_id = 4 -> null
-       user    = "acchs" -> null
    }

  # forgejo_team_member.devops_indigo0445 will be created
+ resource "forgejo_team_member" "devops_indigo0445" {
+       team_id = 4
+       user    = "rjwu"
    }

  # forgejo_team_member.tech_leads_anish will be destroyed
  # (because forgejo_team_member.tech_leads_anish is not in configuration)
- resource "forgejo_team_member" "tech_leads_anish" {
-       team_id = 44 -> null
-       user    = "anish" -> null
    }

  # forgejo_team_member.tech_leads_annadavi will be destroyed
  # (because forgejo_team_member.tech_leads_annadavi is not in configuration)
- resource "forgejo_team_member" "tech_leads_annadavi" {
-       team_id = 44 -> null
-       user    = "annadavi" -> null
    }

  # forgejo_team_member.tech_leads_arom1a will be destroyed
  # (because forgejo_team_member.tech_leads_arom1a is not in configuration)
- resource "forgejo_team_member" "tech_leads_arom1a" {
-       team_id = 44 -> null
-       user    = "arom1a" -> null
    }

  # forgejo_team_member.tech_leads_dcrodrig will be destroyed
  # (because forgejo_team_member.tech_leads_dcrodrig is not in configuration)
- resource "forgejo_team_member" "tech_leads_dcrodrig" {
-       team_id = 44 -> null
-       user    = "dcrodrig" -> null
    }

  # forgejo_team_member.tech_leads_emmawong will be destroyed
  # (because forgejo_team_member.tech_leads_emmawong is not in configuration)
- resource "forgejo_team_member" "tech_leads_emmawong" {
-       team_id = 44 -> null
-       user    = "emmawong" -> null
    }

  # forgejo_team_member.tech_leads_etashjha will be destroyed
  # (because forgejo_team_member.tech_leads_etashjha is not in configuration)
- resource "forgejo_team_member" "tech_leads_etashjha" {
-       team_id = 44 -> null
-       user    = "etashjha" -> null
    }

  # forgejo_team_member.tech_leads_gostmeaper will be destroyed
  # (because forgejo_team_member.tech_leads_gostmeaper is not in configuration)
- resource "forgejo_team_member" "tech_leads_gostmeaper" {
-       team_id = 44 -> null
-       user    = "johncao" -> null
    }

  # forgejo_team_member.tech_leads_jalenluorion will be destroyed
  # (because forgejo_team_member.tech_leads_jalenluorion is not in configuration)
- resource "forgejo_team_member" "tech_leads_jalenluorion" {
-       team_id = 44 -> null
-       user    = "jalenluorion" -> null
    }

  # forgejo_team_member.tech_leads_krishsax will be destroyed
  # (because forgejo_team_member.tech_leads_krishsax is not in configuration)
- resource "forgejo_team_member" "tech_leads_krishsax" {
-       team_id = 44 -> null
-       user    = "krishsax" -> null
    }

  # forgejo_team_member.tech_leads_kritdass will be destroyed
  # (because forgejo_team_member.tech_leads_kritdass is not in configuration)
- resource "forgejo_team_member" "tech_leads_kritdass" {
-       team_id = 44 -> null
-       user    = "kdass" -> null
    }

  # forgejo_team_member.tech_leads_maybe_yiyi will be destroyed
  # (because forgejo_team_member.tech_leads_maybe_yiyi is not in configuration)
- resource "forgejo_team_member" "tech_leads_maybe_yiyi" {
-       team_id = 44 -> null
-       user    = "yiyoungl" -> null
    }

  # forgejo_team_member.tech_leads_nikhilr will be destroyed
  # (because forgejo_team_member.tech_leads_nikhilr is not in configuration)
- resource "forgejo_team_member" "tech_leads_nikhilr" {
-       team_id = 44 -> null
-       user    = "nramanuj" -> null
    }

  # forgejo_team_member.tech_leads_siddharthr will be destroyed
  # (because forgejo_team_member.tech_leads_siddharthr is not in configuration)
- resource "forgejo_team_member" "tech_leads_siddharthr" {
-       team_id = 44 -> null
-       user    = "sradhak2" -> null
    }

  # forgejo_team_member.tech_leads_tentype will be destroyed
  # (because forgejo_team_member.tech_leads_tentype is not in configuration)
- resource "forgejo_team_member" "tech_leads_tentype" {
-       team_id = 44 -> null
-       user    = "maxwen" -> null
    }

  # forgejo_team_member.tech_leads_thesuperrl will be destroyed
  # (because forgejo_team_member.tech_leads_thesuperrl is not in configuration)
- resource "forgejo_team_member" "tech_leads_thesuperrl" {
-       team_id = 44 -> null
-       user    = "thesuperrl" -> null
    }

  # forgejo_team_member.tech_leads_xboxbedrock will be destroyed
  # (because forgejo_team_member.tech_leads_xboxbedrock is not in configuration)
- resource "forgejo_team_member" "tech_leads_xboxbedrock" {
-       team_id = 44 -> null
-       user    = "xboxbedrock" -> null
    }

  # forgejo_team_repository.tech_leads_governance will be destroyed
  # (because forgejo_team_repository.tech_leads_governance is not in configuration)
- resource "forgejo_team_repository" "tech_leads_governance" {
-       owner      = "ScottyLabs" -> null
-       repository = "governance" -> null
-       team_id    = 44 -> null
    }

  # github_repository.slai_cmugpt_surface will be updated in-place
! resource "github_repository" "slai_cmugpt_surface" {
        id                                      = "cmugpt-surface"
        name                                    = "cmugpt-surface"
        # (39 unchanged attributes hidden)

-       template {
-           include_all_branches = false -> null
-           owner                = "scottylabs-labrador" -> null
-           repository           = "ScottyStack" -> null
        }

        # (1 unchanged block hidden)
    }

  # github_team_membership.devops_acchs will be destroyed
  # (because github_team_membership.devops_acchs is not in configuration)
- resource "github_team_membership" "devops_acchs" {
-       etag     = "W/\"4179f054de82f27572bb3c7ea816b563cafc9fbc4d35a96fb398e60b043d44d1\"" -> null
-       id       = "14706218:acchs-at-cmu" -> null
-       role     = "member" -> null
-       team_id  = "14706218" -> null
-       username = "acchs-at-cmu" -> null
    }

  # github_team_membership.devops_indigo0445 will be created
+ resource "github_team_membership" "devops_indigo0445" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "14706218"
+       username = "indigo0445"
    }

Plan: 2 to add, 1 to change, 20 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 1 to change, 20 to destroy.


Plan Summary

6 projects, 5 with changes, 1 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.34s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # null_resource.slack_devops_ch0_acchs will be destroyed # (because null_resource.slack_devops_ch0_acchs is not in configuration) - resource "null_resource" "slack_devops_ch0_acchs" { - id = "5494655407757332457" -> null - triggers = { - "channel" = "C08K3Q77ZQF" - "user" = "U0BVBNX8YGZ" } -> null } # null_resource.slack_devops_ch0_indigo0445 will be created + resource "null_resource" "slack_devops_ch0_indigo0445" { + id = (known after apply) + triggers = { + "channel" = "C08K3Q77ZQF" + "user" = "U09C4RTR36F" } } # null_resource.slack_devops_ch1_acchs will be destroyed # (because null_resource.slack_devops_ch1_acchs is not in configuration) - resource "null_resource" "slack_devops_ch1_acchs" { - id = "1105780819490615240" -> null - triggers = { - "channel" = "C0BNV84NUTC" - "user" = "U0BVBNX8YGZ" } -> null } # null_resource.slack_devops_ch1_indigo0445 will be created + resource "null_resource" "slack_devops_ch1_indigo0445" { + id = (known after apply) + triggers = { + "channel" = "C0BNV84NUTC" + "user" = "U09C4RTR36F" } } # null_resource.slack_hub_indigo0445 will be created + resource "null_resource" "slack_hub_indigo0445" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U09C4RTR36F" } } Plan: 3 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 3 to add, 0 to change, 2 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.29s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on shared package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.39s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_indigo0445 will be created + resource "google_cloud_identity_group_membership" "tech_indigo0445" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "rjwu@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.31s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_devops will be updated in-place ! resource "keycloak_group_memberships" "project_devops" { id = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7" ! members = [ - "azzhou", + "rjwu", # (3 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_changelog will be updated in-place ! resource "keycloak_group_memberships" "project_devops_changelog" { id = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77" ! members = [ - "azzhou", + "rjwu", # (3 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_documentation will be updated in-place ! resource "keycloak_group_memberships" "project_devops_documentation" { id = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f" ! members = [ - "azzhou", + "rjwu", # (3 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_link-shortener will be updated in-place ! resource "keycloak_group_memberships" "project_devops_link-shortener" { id = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935" ! members = [ - "azzhou", + "rjwu", # (3 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "rjwu", # (36 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 5 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 5 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.32s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # discord_member_roles.acchs will be updated in-place ! resource "discord_member_roles" "acchs" { id = "1353118209024331948:781930226778767380" # (2 unchanged attributes hidden) - role { - has_role = true -> null - role_id = "1488787087598813215" -> null } # (2 unchanged blocks hidden) } # discord_member_roles.indigo0445 will be created + resource "discord_member_roles" "indigo0445" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "629857326232371210" + role { + has_role = true + role_id = "1488787087598813215" } + role { + has_role = true + role_id = "1488787088118644848" } } Plan: 1 to add, 1 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 1 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.39s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) - destroy OpenTofu will perform the following actions: # forgejo_team.tech_leads will be destroyed # (because forgejo_team.tech_leads is not in configuration) - resource "forgejo_team" "tech_leads" { - can_create_org_repo = false -> null - description = "All tech leads (write access to governance)" -> null - id = 44 -> null - includes_all_repositories = false -> null - name = "tech-leads" -> null - organization = "ScottyLabs" -> null - organization_id = 3 -> null - permission = "write" -> null - units_map = { - "repo.code" = "write" - "repo.issues" = "write" - "repo.pulls" = "write" } -> null } # forgejo_team_member.devops_acchs will be destroyed # (because forgejo_team_member.devops_acchs is not in configuration) - resource "forgejo_team_member" "devops_acchs" { - team_id = 4 -> null - user = "acchs" -> null } # forgejo_team_member.devops_indigo0445 will be created + resource "forgejo_team_member" "devops_indigo0445" { + team_id = 4 + user = "rjwu" } # forgejo_team_member.tech_leads_anish will be destroyed # (because forgejo_team_member.tech_leads_anish is not in configuration) - resource "forgejo_team_member" "tech_leads_anish" { - team_id = 44 -> null - user = "anish" -> null } # forgejo_team_member.tech_leads_annadavi will be destroyed # (because forgejo_team_member.tech_leads_annadavi is not in configuration) - resource "forgejo_team_member" "tech_leads_annadavi" { - team_id = 44 -> null - user = "annadavi" -> null } # forgejo_team_member.tech_leads_arom1a will be destroyed # (because forgejo_team_member.tech_leads_arom1a is not in configuration) - resource "forgejo_team_member" "tech_leads_arom1a" { - team_id = 44 -> null - user = "arom1a" -> null } # forgejo_team_member.tech_leads_dcrodrig will be destroyed # (because forgejo_team_member.tech_leads_dcrodrig is not in configuration) - resource "forgejo_team_member" "tech_leads_dcrodrig" { - team_id = 44 -> null - user = "dcrodrig" -> null } # forgejo_team_member.tech_leads_emmawong will be destroyed # (because forgejo_team_member.tech_leads_emmawong is not in configuration) - resource "forgejo_team_member" "tech_leads_emmawong" { - team_id = 44 -> null - user = "emmawong" -> null } # forgejo_team_member.tech_leads_etashjha will be destroyed # (because forgejo_team_member.tech_leads_etashjha is not in configuration) - resource "forgejo_team_member" "tech_leads_etashjha" { - team_id = 44 -> null - user = "etashjha" -> null } # forgejo_team_member.tech_leads_gostmeaper will be destroyed # (because forgejo_team_member.tech_leads_gostmeaper is not in configuration) - resource "forgejo_team_member" "tech_leads_gostmeaper" { - team_id = 44 -> null - user = "johncao" -> null } # forgejo_team_member.tech_leads_jalenluorion will be destroyed # (because forgejo_team_member.tech_leads_jalenluorion is not in configuration) - resource "forgejo_team_member" "tech_leads_jalenluorion" { - team_id = 44 -> null - user = "jalenluorion" -> null } # forgejo_team_member.tech_leads_krishsax will be destroyed # (because forgejo_team_member.tech_leads_krishsax is not in configuration) - resource "forgejo_team_member" "tech_leads_krishsax" { - team_id = 44 -> null - user = "krishsax" -> null } # forgejo_team_member.tech_leads_kritdass will be destroyed # (because forgejo_team_member.tech_leads_kritdass is not in configuration) - resource "forgejo_team_member" "tech_leads_kritdass" { - team_id = 44 -> null - user = "kdass" -> null } # forgejo_team_member.tech_leads_maybe_yiyi will be destroyed # (because forgejo_team_member.tech_leads_maybe_yiyi is not in configuration) - resource "forgejo_team_member" "tech_leads_maybe_yiyi" { - team_id = 44 -> null - user = "yiyoungl" -> null } # forgejo_team_member.tech_leads_nikhilr will be destroyed # (because forgejo_team_member.tech_leads_nikhilr is not in configuration) - resource "forgejo_team_member" "tech_leads_nikhilr" { - team_id = 44 -> null - user = "nramanuj" -> null } # forgejo_team_member.tech_leads_siddharthr will be destroyed # (because forgejo_team_member.tech_leads_siddharthr is not in configuration) - resource "forgejo_team_member" "tech_leads_siddharthr" { - team_id = 44 -> null - user = "sradhak2" -> null } # forgejo_team_member.tech_leads_tentype will be destroyed # (because forgejo_team_member.tech_leads_tentype is not in configuration) - resource "forgejo_team_member" "tech_leads_tentype" { - team_id = 44 -> null - user = "maxwen" -> null } # forgejo_team_member.tech_leads_thesuperrl will be destroyed # (because forgejo_team_member.tech_leads_thesuperrl is not in configuration) - resource "forgejo_team_member" "tech_leads_thesuperrl" { - team_id = 44 -> null - user = "thesuperrl" -> null } # forgejo_team_member.tech_leads_xboxbedrock will be destroyed # (because forgejo_team_member.tech_leads_xboxbedrock is not in configuration) - resource "forgejo_team_member" "tech_leads_xboxbedrock" { - team_id = 44 -> null - user = "xboxbedrock" -> null } # forgejo_team_repository.tech_leads_governance will be destroyed # (because forgejo_team_repository.tech_leads_governance is not in configuration) - resource "forgejo_team_repository" "tech_leads_governance" { - owner = "ScottyLabs" -> null - repository = "governance" -> null - team_id = 44 -> null } # github_repository.slai_cmugpt_surface will be updated in-place ! resource "github_repository" "slai_cmugpt_surface" { id = "cmugpt-surface" name = "cmugpt-surface" # (39 unchanged attributes hidden) - template { - include_all_branches = false -> null - owner = "scottylabs-labrador" -> null - repository = "ScottyStack" -> null } # (1 unchanged block hidden) } # github_team_membership.devops_acchs will be destroyed # (because github_team_membership.devops_acchs is not in configuration) - resource "github_team_membership" "devops_acchs" { - etag = "W/\"4179f054de82f27572bb3c7ea816b563cafc9fbc4d35a96fb398e60b043d44d1\"" -> null - id = "14706218:acchs-at-cmu" -> null - role = "member" -> null - team_id = "14706218" -> null - username = "acchs-at-cmu" -> null } # github_team_membership.devops_indigo0445 will be created + resource "github_team_membership" "devops_indigo0445" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "14706218" + username = "indigo0445" } Plan: 2 to add, 1 to change, 20 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 1 to change, 20 to destroy. --- ### Plan Summary 6 projects, 5 with changes, 1 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis unlock

atlantis unlock

All Atlantis locks for this PR have been unlocked and plans discarded

All Atlantis locks for this PR have been unlocked and plans discarded
Member

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.13s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_devops_ch0_acchs will be destroyed
  # (because null_resource.slack_devops_ch0_acchs is not in configuration)
- resource "null_resource" "slack_devops_ch0_acchs" {
-       id       = "5494655407757332457" -> null
-       triggers = {
          - "channel" = "C08K3Q77ZQF"
          - "user"    = "U0BVBNX8YGZ"
        } -> null
    }

  # null_resource.slack_devops_ch0_aszhou will be destroyed
  # (because null_resource.slack_devops_ch0_aszhou is not in configuration)
- resource "null_resource" "slack_devops_ch0_aszhou" {
-       id       = "528167437641880870" -> null
-       triggers = {
          - "channel" = "C08K3Q77ZQF"
          - "user"    = "U09CQ04V4BT"
        } -> null
    }

  # null_resource.slack_devops_ch0_indigo0445 will be created
+ resource "null_resource" "slack_devops_ch0_indigo0445" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08K3Q77ZQF"
          + "user"    = "U09C4RTR36F"
        }
    }

  # null_resource.slack_devops_ch1_acchs will be destroyed
  # (because null_resource.slack_devops_ch1_acchs is not in configuration)
- resource "null_resource" "slack_devops_ch1_acchs" {
-       id       = "1105780819490615240" -> null
-       triggers = {
          - "channel" = "C0BNV84NUTC"
          - "user"    = "U0BVBNX8YGZ"
        } -> null
    }

  # null_resource.slack_devops_ch1_aszhou will be destroyed
  # (because null_resource.slack_devops_ch1_aszhou is not in configuration)
- resource "null_resource" "slack_devops_ch1_aszhou" {
-       id       = "2598782963452480140" -> null
-       triggers = {
          - "channel" = "C0BNV84NUTC"
          - "user"    = "U09CQ04V4BT"
        } -> null
    }

  # null_resource.slack_devops_ch1_indigo0445 will be created
+ resource "null_resource" "slack_devops_ch1_indigo0445" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0BNV84NUTC"
          + "user"    = "U09C4RTR36F"
        }
    }

  # null_resource.slack_hub_aszhou will be destroyed
  # (because null_resource.slack_hub_aszhou is not in configuration)
- resource "null_resource" "slack_hub_aszhou" {
-       id       = "7120216021093146134" -> null
-       triggers = {
          - "channel" = "C096TM8EMS8"
          - "user"    = "U09CQ04V4BT"
        } -> null
    }

  # null_resource.slack_hub_indigo0445 will be created
+ resource "null_resource" "slack_hub_indigo0445" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U09C4RTR36F"
        }
    }

Plan: 3 to add, 0 to change, 5 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 3 to add, 0 to change, 5 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.19s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.20s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_aszhou will be destroyed
  # (because google_cloud_identity_group_membership.tech_aszhou is not in configuration)
- resource "google_cloud_identity_group_membership" "tech_aszhou" {
-       create_time = "2026-09-19T20:39:12.556017Z" -> null
-       group       = "groups/04anzqyu1bqbv79" -> null
-       id          = "groups/04anzqyu1bqbv79/memberships/100575886719005530375" -> null
-       name        = "groups/04anzqyu1bqbv79/memberships/100575886719005530375" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-19T20:39:12.556017Z" -> null

-       preferred_member_key {
-           id = "aszhou@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.tech_indigo0445 will be created
+ resource "google_cloud_identity_group_membership" "tech_indigo0445" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "rjwu@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 1 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 1 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.22s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_devops will be updated in-place
! resource "keycloak_group_memberships" "project_devops" {
        id       = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7"
!       members  = [
-           "aszhou",
-           "azzhou",
+           "rjwu",
            # (3 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_changelog will be updated in-place
! resource "keycloak_group_memberships" "project_devops_changelog" {
        id       = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77"
!       members  = [
-           "aszhou",
-           "azzhou",
+           "rjwu",
            # (3 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_documentation will be updated in-place
! resource "keycloak_group_memberships" "project_devops_documentation" {
        id       = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f"
!       members  = [
-           "aszhou",
-           "azzhou",
+           "rjwu",
            # (3 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_link-shortener will be updated in-place
! resource "keycloak_group_memberships" "project_devops_link-shortener" {
        id       = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935"
!       members  = [
-           "aszhou",
-           "azzhou",
+           "rjwu",
            # (3 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
-           "aszhou",
+           "rjwu",
            # (36 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 5 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 5 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.22s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)
- destroy

OpenTofu will perform the following actions:

  # discord_member_roles.acchs will be updated in-place
! resource "discord_member_roles" "acchs" {
        id        = "1353118209024331948:781930226778767380"
        # (2 unchanged attributes hidden)

-       role {
-           has_role = true -> null
-           role_id  = "1488787087598813215" -> null
        }

        # (2 unchanged blocks hidden)
    }

  # discord_member_roles.aszhou will be destroyed
  # (because discord_member_roles.aszhou is not in configuration)
- resource "discord_member_roles" "aszhou" {
-       id        = "1353118209024331948:572567442153406479" -> null
-       server_id = "1353118209024331948" -> null
-       user_id   = "572567442153406479" -> null

-       role {
-           has_role = true -> null
-           role_id  = "1488787087598813215" -> null
        }
-       role {
-           has_role = true -> null
-           role_id  = "1488787088118644848" -> null
        }
    }

  # discord_member_roles.indigo0445 will be created
+ resource "discord_member_roles" "indigo0445" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "629857326232371210"

+       role {
+           has_role = true
+           role_id  = "1488787087598813215"
        }
+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
    }

Plan: 1 to add, 1 to change, 1 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 1 to change, 1 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.15s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)
- destroy

OpenTofu will perform the following actions:

  # forgejo_team.tech_leads will be destroyed
  # (because forgejo_team.tech_leads is not in configuration)
- resource "forgejo_team" "tech_leads" {
-       can_create_org_repo       = false -> null
-       description               = "All tech leads (write access to governance)" -> null
-       id                        = 44 -> null
-       includes_all_repositories = false -> null
-       name                      = "tech-leads" -> null
-       organization              = "ScottyLabs" -> null
-       organization_id           = 3 -> null
-       permission                = "write" -> null
-       units_map                 = {
          - "repo.code"   = "write"
          - "repo.issues" = "write"
          - "repo.pulls"  = "write"
        } -> null
    }

  # forgejo_team_member.devops_acchs will be destroyed
  # (because forgejo_team_member.devops_acchs is not in configuration)
- resource "forgejo_team_member" "devops_acchs" {
-       team_id = 4 -> null
-       user    = "acchs" -> null
    }

  # forgejo_team_member.devops_aszhou will be destroyed
  # (because forgejo_team_member.devops_aszhou is not in configuration)
- resource "forgejo_team_member" "devops_aszhou" {
-       team_id = 4 -> null
-       user    = "aszhou" -> null
    }

  # forgejo_team_member.devops_indigo0445 will be created
+ resource "forgejo_team_member" "devops_indigo0445" {
+       team_id = 4
+       user    = "rjwu"
    }

  # forgejo_team_member.tech_leads_anish will be destroyed
  # (because forgejo_team_member.tech_leads_anish is not in configuration)
- resource "forgejo_team_member" "tech_leads_anish" {
-       team_id = 44 -> null
-       user    = "anish" -> null
    }

  # forgejo_team_member.tech_leads_annadavi will be destroyed
  # (because forgejo_team_member.tech_leads_annadavi is not in configuration)
- resource "forgejo_team_member" "tech_leads_annadavi" {
-       team_id = 44 -> null
-       user    = "annadavi" -> null
    }

  # forgejo_team_member.tech_leads_arom1a will be destroyed
  # (because forgejo_team_member.tech_leads_arom1a is not in configuration)
- resource "forgejo_team_member" "tech_leads_arom1a" {
-       team_id = 44 -> null
-       user    = "arom1a" -> null
    }

  # forgejo_team_member.tech_leads_dcrodrig will be destroyed
  # (because forgejo_team_member.tech_leads_dcrodrig is not in configuration)
- resource "forgejo_team_member" "tech_leads_dcrodrig" {
-       team_id = 44 -> null
-       user    = "dcrodrig" -> null
    }

  # forgejo_team_member.tech_leads_emmawong will be destroyed
  # (because forgejo_team_member.tech_leads_emmawong is not in configuration)
- resource "forgejo_team_member" "tech_leads_emmawong" {
-       team_id = 44 -> null
-       user    = "emmawong" -> null
    }

  # forgejo_team_member.tech_leads_etashjha will be destroyed
  # (because forgejo_team_member.tech_leads_etashjha is not in configuration)
- resource "forgejo_team_member" "tech_leads_etashjha" {
-       team_id = 44 -> null
-       user    = "etashjha" -> null
    }

  # forgejo_team_member.tech_leads_gostmeaper will be destroyed
  # (because forgejo_team_member.tech_leads_gostmeaper is not in configuration)
- resource "forgejo_team_member" "tech_leads_gostmeaper" {
-       team_id = 44 -> null
-       user    = "johncao" -> null
    }

  # forgejo_team_member.tech_leads_jalenluorion will be destroyed
  # (because forgejo_team_member.tech_leads_jalenluorion is not in configuration)
- resource "forgejo_team_member" "tech_leads_jalenluorion" {
-       team_id = 44 -> null
-       user    = "jalenluorion" -> null
    }

  # forgejo_team_member.tech_leads_krishsax will be destroyed
  # (because forgejo_team_member.tech_leads_krishsax is not in configuration)
- resource "forgejo_team_member" "tech_leads_krishsax" {
-       team_id = 44 -> null
-       user    = "krishsax" -> null
    }

  # forgejo_team_member.tech_leads_kritdass will be destroyed
  # (because forgejo_team_member.tech_leads_kritdass is not in configuration)
- resource "forgejo_team_member" "tech_leads_kritdass" {
-       team_id = 44 -> null
-       user    = "kdass" -> null
    }

  # forgejo_team_member.tech_leads_maybe_yiyi will be destroyed
  # (because forgejo_team_member.tech_leads_maybe_yiyi is not in configuration)
- resource "forgejo_team_member" "tech_leads_maybe_yiyi" {
-       team_id = 44 -> null
-       user    = "yiyoungl" -> null
    }

  # forgejo_team_member.tech_leads_nikhilr will be destroyed
  # (because forgejo_team_member.tech_leads_nikhilr is not in configuration)
- resource "forgejo_team_member" "tech_leads_nikhilr" {
-       team_id = 44 -> null
-       user    = "nramanuj" -> null
    }

  # forgejo_team_member.tech_leads_siddharthr will be destroyed
  # (because forgejo_team_member.tech_leads_siddharthr is not in configuration)
- resource "forgejo_team_member" "tech_leads_siddharthr" {
-       team_id = 44 -> null
-       user    = "sradhak2" -> null
    }

  # forgejo_team_member.tech_leads_tentype will be destroyed
  # (because forgejo_team_member.tech_leads_tentype is not in configuration)
- resource "forgejo_team_member" "tech_leads_tentype" {
-       team_id = 44 -> null
-       user    = "maxwen" -> null
    }

  # forgejo_team_member.tech_leads_thesuperrl will be destroyed
  # (because forgejo_team_member.tech_leads_thesuperrl is not in configuration)
- resource "forgejo_team_member" "tech_leads_thesuperrl" {
-       team_id = 44 -> null
-       user    = "thesuperrl" -> null
    }

  # forgejo_team_member.tech_leads_xboxbedrock will be destroyed
  # (because forgejo_team_member.tech_leads_xboxbedrock is not in configuration)
- resource "forgejo_team_member" "tech_leads_xboxbedrock" {
-       team_id = 44 -> null
-       user    = "xboxbedrock" -> null
    }

  # forgejo_team_repository.tech_leads_governance will be destroyed
  # (because forgejo_team_repository.tech_leads_governance is not in configuration)
- resource "forgejo_team_repository" "tech_leads_governance" {
-       owner      = "ScottyLabs" -> null
-       repository = "governance" -> null
-       team_id    = 44 -> null
    }

  # github_repository.slai_cmugpt_surface will be updated in-place
! resource "github_repository" "slai_cmugpt_surface" {
        id                                      = "cmugpt-surface"
        name                                    = "cmugpt-surface"
        # (39 unchanged attributes hidden)

-       template {
-           include_all_branches = false -> null
-           owner                = "scottylabs-labrador" -> null
-           repository           = "ScottyStack" -> null
        }

        # (1 unchanged block hidden)
    }

  # github_team_membership.devops_acchs will be destroyed
  # (because github_team_membership.devops_acchs is not in configuration)
- resource "github_team_membership" "devops_acchs" {
-       etag     = "W/\"4179f054de82f27572bb3c7ea816b563cafc9fbc4d35a96fb398e60b043d44d1\"" -> null
-       id       = "14706218:acchs-at-cmu" -> null
-       role     = "member" -> null
-       team_id  = "14706218" -> null
-       username = "acchs-at-cmu" -> null
    }

  # github_team_membership.devops_aszhou will be destroyed
  # (because github_team_membership.devops_aszhou is not in configuration)
- resource "github_team_membership" "devops_aszhou" {
-       etag     = "W/\"7ded4c758e7d683f26bbd6ba5249b0169eb642c18bd932c07b551232053c34d8\"" -> null
-       id       = "14706218:azhou22" -> null
-       role     = "member" -> null
-       team_id  = "14706218" -> null
-       username = "azhou22" -> null
    }

  # github_team_membership.devops_indigo0445 will be created
+ resource "github_team_membership" "devops_indigo0445" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "14706218"
+       username = "indigo0445"
    }

Plan: 2 to add, 1 to change, 22 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 1 to change, 22 to destroy.


Plan Summary

6 projects, 5 with changes, 1 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.13s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # null_resource.slack_devops_ch0_acchs will be destroyed # (because null_resource.slack_devops_ch0_acchs is not in configuration) - resource "null_resource" "slack_devops_ch0_acchs" { - id = "5494655407757332457" -> null - triggers = { - "channel" = "C08K3Q77ZQF" - "user" = "U0BVBNX8YGZ" } -> null } # null_resource.slack_devops_ch0_aszhou will be destroyed # (because null_resource.slack_devops_ch0_aszhou is not in configuration) - resource "null_resource" "slack_devops_ch0_aszhou" { - id = "528167437641880870" -> null - triggers = { - "channel" = "C08K3Q77ZQF" - "user" = "U09CQ04V4BT" } -> null } # null_resource.slack_devops_ch0_indigo0445 will be created + resource "null_resource" "slack_devops_ch0_indigo0445" { + id = (known after apply) + triggers = { + "channel" = "C08K3Q77ZQF" + "user" = "U09C4RTR36F" } } # null_resource.slack_devops_ch1_acchs will be destroyed # (because null_resource.slack_devops_ch1_acchs is not in configuration) - resource "null_resource" "slack_devops_ch1_acchs" { - id = "1105780819490615240" -> null - triggers = { - "channel" = "C0BNV84NUTC" - "user" = "U0BVBNX8YGZ" } -> null } # null_resource.slack_devops_ch1_aszhou will be destroyed # (because null_resource.slack_devops_ch1_aszhou is not in configuration) - resource "null_resource" "slack_devops_ch1_aszhou" { - id = "2598782963452480140" -> null - triggers = { - "channel" = "C0BNV84NUTC" - "user" = "U09CQ04V4BT" } -> null } # null_resource.slack_devops_ch1_indigo0445 will be created + resource "null_resource" "slack_devops_ch1_indigo0445" { + id = (known after apply) + triggers = { + "channel" = "C0BNV84NUTC" + "user" = "U09C4RTR36F" } } # null_resource.slack_hub_aszhou will be destroyed # (because null_resource.slack_hub_aszhou is not in configuration) - resource "null_resource" "slack_hub_aszhou" { - id = "7120216021093146134" -> null - triggers = { - "channel" = "C096TM8EMS8" - "user" = "U09CQ04V4BT" } -> null } # null_resource.slack_hub_indigo0445 will be created + resource "null_resource" "slack_hub_indigo0445" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U09C4RTR36F" } } Plan: 3 to add, 0 to change, 5 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 3 to add, 0 to change, 5 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.19s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.20s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_aszhou will be destroyed # (because google_cloud_identity_group_membership.tech_aszhou is not in configuration) - resource "google_cloud_identity_group_membership" "tech_aszhou" { - create_time = "2026-09-19T20:39:12.556017Z" -> null - group = "groups/04anzqyu1bqbv79" -> null - id = "groups/04anzqyu1bqbv79/memberships/100575886719005530375" -> null - name = "groups/04anzqyu1bqbv79/memberships/100575886719005530375" -> null - type = "USER" -> null - update_time = "2026-09-19T20:39:12.556017Z" -> null - preferred_member_key { - id = "aszhou@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.tech_indigo0445 will be created + resource "google_cloud_identity_group_membership" "tech_indigo0445" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "rjwu@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 1 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 1 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.22s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_devops will be updated in-place ! resource "keycloak_group_memberships" "project_devops" { id = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7" ! members = [ - "aszhou", - "azzhou", + "rjwu", # (3 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_changelog will be updated in-place ! resource "keycloak_group_memberships" "project_devops_changelog" { id = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77" ! members = [ - "aszhou", - "azzhou", + "rjwu", # (3 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_documentation will be updated in-place ! resource "keycloak_group_memberships" "project_devops_documentation" { id = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f" ! members = [ - "aszhou", - "azzhou", + "rjwu", # (3 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_link-shortener will be updated in-place ! resource "keycloak_group_memberships" "project_devops_link-shortener" { id = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935" ! members = [ - "aszhou", - "azzhou", + "rjwu", # (3 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ - "aszhou", + "rjwu", # (36 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 5 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 5 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.22s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) - destroy OpenTofu will perform the following actions: # discord_member_roles.acchs will be updated in-place ! resource "discord_member_roles" "acchs" { id = "1353118209024331948:781930226778767380" # (2 unchanged attributes hidden) - role { - has_role = true -> null - role_id = "1488787087598813215" -> null } # (2 unchanged blocks hidden) } # discord_member_roles.aszhou will be destroyed # (because discord_member_roles.aszhou is not in configuration) - resource "discord_member_roles" "aszhou" { - id = "1353118209024331948:572567442153406479" -> null - server_id = "1353118209024331948" -> null - user_id = "572567442153406479" -> null - role { - has_role = true -> null - role_id = "1488787087598813215" -> null } - role { - has_role = true -> null - role_id = "1488787088118644848" -> null } } # discord_member_roles.indigo0445 will be created + resource "discord_member_roles" "indigo0445" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "629857326232371210" + role { + has_role = true + role_id = "1488787087598813215" } + role { + has_role = true + role_id = "1488787088118644848" } } Plan: 1 to add, 1 to change, 1 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 1 to change, 1 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.15s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) - destroy OpenTofu will perform the following actions: # forgejo_team.tech_leads will be destroyed # (because forgejo_team.tech_leads is not in configuration) - resource "forgejo_team" "tech_leads" { - can_create_org_repo = false -> null - description = "All tech leads (write access to governance)" -> null - id = 44 -> null - includes_all_repositories = false -> null - name = "tech-leads" -> null - organization = "ScottyLabs" -> null - organization_id = 3 -> null - permission = "write" -> null - units_map = { - "repo.code" = "write" - "repo.issues" = "write" - "repo.pulls" = "write" } -> null } # forgejo_team_member.devops_acchs will be destroyed # (because forgejo_team_member.devops_acchs is not in configuration) - resource "forgejo_team_member" "devops_acchs" { - team_id = 4 -> null - user = "acchs" -> null } # forgejo_team_member.devops_aszhou will be destroyed # (because forgejo_team_member.devops_aszhou is not in configuration) - resource "forgejo_team_member" "devops_aszhou" { - team_id = 4 -> null - user = "aszhou" -> null } # forgejo_team_member.devops_indigo0445 will be created + resource "forgejo_team_member" "devops_indigo0445" { + team_id = 4 + user = "rjwu" } # forgejo_team_member.tech_leads_anish will be destroyed # (because forgejo_team_member.tech_leads_anish is not in configuration) - resource "forgejo_team_member" "tech_leads_anish" { - team_id = 44 -> null - user = "anish" -> null } # forgejo_team_member.tech_leads_annadavi will be destroyed # (because forgejo_team_member.tech_leads_annadavi is not in configuration) - resource "forgejo_team_member" "tech_leads_annadavi" { - team_id = 44 -> null - user = "annadavi" -> null } # forgejo_team_member.tech_leads_arom1a will be destroyed # (because forgejo_team_member.tech_leads_arom1a is not in configuration) - resource "forgejo_team_member" "tech_leads_arom1a" { - team_id = 44 -> null - user = "arom1a" -> null } # forgejo_team_member.tech_leads_dcrodrig will be destroyed # (because forgejo_team_member.tech_leads_dcrodrig is not in configuration) - resource "forgejo_team_member" "tech_leads_dcrodrig" { - team_id = 44 -> null - user = "dcrodrig" -> null } # forgejo_team_member.tech_leads_emmawong will be destroyed # (because forgejo_team_member.tech_leads_emmawong is not in configuration) - resource "forgejo_team_member" "tech_leads_emmawong" { - team_id = 44 -> null - user = "emmawong" -> null } # forgejo_team_member.tech_leads_etashjha will be destroyed # (because forgejo_team_member.tech_leads_etashjha is not in configuration) - resource "forgejo_team_member" "tech_leads_etashjha" { - team_id = 44 -> null - user = "etashjha" -> null } # forgejo_team_member.tech_leads_gostmeaper will be destroyed # (because forgejo_team_member.tech_leads_gostmeaper is not in configuration) - resource "forgejo_team_member" "tech_leads_gostmeaper" { - team_id = 44 -> null - user = "johncao" -> null } # forgejo_team_member.tech_leads_jalenluorion will be destroyed # (because forgejo_team_member.tech_leads_jalenluorion is not in configuration) - resource "forgejo_team_member" "tech_leads_jalenluorion" { - team_id = 44 -> null - user = "jalenluorion" -> null } # forgejo_team_member.tech_leads_krishsax will be destroyed # (because forgejo_team_member.tech_leads_krishsax is not in configuration) - resource "forgejo_team_member" "tech_leads_krishsax" { - team_id = 44 -> null - user = "krishsax" -> null } # forgejo_team_member.tech_leads_kritdass will be destroyed # (because forgejo_team_member.tech_leads_kritdass is not in configuration) - resource "forgejo_team_member" "tech_leads_kritdass" { - team_id = 44 -> null - user = "kdass" -> null } # forgejo_team_member.tech_leads_maybe_yiyi will be destroyed # (because forgejo_team_member.tech_leads_maybe_yiyi is not in configuration) - resource "forgejo_team_member" "tech_leads_maybe_yiyi" { - team_id = 44 -> null - user = "yiyoungl" -> null } # forgejo_team_member.tech_leads_nikhilr will be destroyed # (because forgejo_team_member.tech_leads_nikhilr is not in configuration) - resource "forgejo_team_member" "tech_leads_nikhilr" { - team_id = 44 -> null - user = "nramanuj" -> null } # forgejo_team_member.tech_leads_siddharthr will be destroyed # (because forgejo_team_member.tech_leads_siddharthr is not in configuration) - resource "forgejo_team_member" "tech_leads_siddharthr" { - team_id = 44 -> null - user = "sradhak2" -> null } # forgejo_team_member.tech_leads_tentype will be destroyed # (because forgejo_team_member.tech_leads_tentype is not in configuration) - resource "forgejo_team_member" "tech_leads_tentype" { - team_id = 44 -> null - user = "maxwen" -> null } # forgejo_team_member.tech_leads_thesuperrl will be destroyed # (because forgejo_team_member.tech_leads_thesuperrl is not in configuration) - resource "forgejo_team_member" "tech_leads_thesuperrl" { - team_id = 44 -> null - user = "thesuperrl" -> null } # forgejo_team_member.tech_leads_xboxbedrock will be destroyed # (because forgejo_team_member.tech_leads_xboxbedrock is not in configuration) - resource "forgejo_team_member" "tech_leads_xboxbedrock" { - team_id = 44 -> null - user = "xboxbedrock" -> null } # forgejo_team_repository.tech_leads_governance will be destroyed # (because forgejo_team_repository.tech_leads_governance is not in configuration) - resource "forgejo_team_repository" "tech_leads_governance" { - owner = "ScottyLabs" -> null - repository = "governance" -> null - team_id = 44 -> null } # github_repository.slai_cmugpt_surface will be updated in-place ! resource "github_repository" "slai_cmugpt_surface" { id = "cmugpt-surface" name = "cmugpt-surface" # (39 unchanged attributes hidden) - template { - include_all_branches = false -> null - owner = "scottylabs-labrador" -> null - repository = "ScottyStack" -> null } # (1 unchanged block hidden) } # github_team_membership.devops_acchs will be destroyed # (because github_team_membership.devops_acchs is not in configuration) - resource "github_team_membership" "devops_acchs" { - etag = "W/\"4179f054de82f27572bb3c7ea816b563cafc9fbc4d35a96fb398e60b043d44d1\"" -> null - id = "14706218:acchs-at-cmu" -> null - role = "member" -> null - team_id = "14706218" -> null - username = "acchs-at-cmu" -> null } # github_team_membership.devops_aszhou will be destroyed # (because github_team_membership.devops_aszhou is not in configuration) - resource "github_team_membership" "devops_aszhou" { - etag = "W/\"7ded4c758e7d683f26bbd6ba5249b0169eb642c18bd932c07b551232053c34d8\"" -> null - id = "14706218:azhou22" -> null - role = "member" -> null - team_id = "14706218" -> null - username = "azhou22" -> null } # github_team_membership.devops_indigo0445 will be created + resource "github_team_membership" "devops_indigo0445" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "14706218" + username = "indigo0445" } Plan: 2 to add, 1 to change, 22 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 1 to change, 22 to destroy. --- ### Plan Summary 6 projects, 5 with changes, 1 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
xboxbedrock force-pushed add-indigo0445-member from bbc2bd3525
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: slack Plan: 3 to add, 0 to change, 5 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 1 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 5 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 1 to change, 1 to destroy.
atlantis/plan: git Plan: 2 to add, 1 to change, 22 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
to a993bc9cce
All checks were successful
CI / build (pull_request) Has been skipped
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 3 to add, 0 to change, 0 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 5 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 2 to add, 1 to change, 0 to destroy.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan 6/6 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
Validate / Governance (pull_request_target) Successful in 9m24s
CI / check-1 (pull_request) Successful in 14m2s
CI / check (pull_request) Successful in 0s
2026-09-19 20:48:46 +00:00
Compare
Member

atlantis plan

atlantis plan

Plan Error

cannot run "plan": pull request 226 is currently locked by "plan" for commit a993bc9.
Wait until the previous command is complete and try again
**Plan Error** ``` cannot run "plan": pull request 226 is currently locked by "plan" for commit a993bc9. Wait until the previous command is complete and try again ``` - Commit: [a993bc9cce24fb70267cd0c45e599cee011e31a9](https://git.cmu.dev/ScottyLabs/governance/pulls/226/commits/a993bc9cce24fb70267cd0c45e599cee011e31a9) - Multiple Atlantis jobs are currently running for this commit.

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.35s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_devops_ch0_indigo0445 will be created
+ resource "null_resource" "slack_devops_ch0_indigo0445" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08K3Q77ZQF"
          + "user"    = "U09C4RTR36F"
        }
    }

  # null_resource.slack_devops_ch1_indigo0445 will be created
+ resource "null_resource" "slack_devops_ch1_indigo0445" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0BNV84NUTC"
          + "user"    = "U09C4RTR36F"
        }
    }

  # null_resource.slack_hub_indigo0445 will be created
+ resource "null_resource" "slack_hub_indigo0445" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U09C4RTR36F"
        }
    }

Plan: 3 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 3 to add, 0 to change, 0 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.32s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/226/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/226/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/226/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 7.32s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_indigo0445 will be created
+ resource "google_cloud_identity_group_membership" "tech_indigo0445" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "rjwu@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.26s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_devops will be updated in-place
! resource "keycloak_group_memberships" "project_devops" {
        id       = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7"
!       members  = [
+           "rjwu",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_changelog will be updated in-place
! resource "keycloak_group_memberships" "project_devops_changelog" {
        id       = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77"
!       members  = [
+           "rjwu",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_documentation will be updated in-place
! resource "keycloak_group_memberships" "project_devops_documentation" {
        id       = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f"
!       members  = [
+           "rjwu",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_link-shortener will be updated in-place
! resource "keycloak_group_memberships" "project_devops_link-shortener" {
        id       = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935"
!       members  = [
+           "rjwu",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "rjwu",
            # (37 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 5 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 5 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.30s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.indigo0445 will be created
+ resource "discord_member_roles" "indigo0445" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "629857326232371210"

+       role {
+           has_role = true
+           role_id  = "1488787087598813215"
        }
+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.30s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # forgejo_team_member.devops_indigo0445 will be created
+ resource "forgejo_team_member" "devops_indigo0445" {
+       team_id = 4
+       user    = "rjwu"
    }

  # github_repository.slai_cmugpt_surface will be updated in-place
! resource "github_repository" "slai_cmugpt_surface" {
        id                                      = "cmugpt-surface"
        name                                    = "cmugpt-surface"
        # (39 unchanged attributes hidden)

-       template {
-           include_all_branches = false -> null
-           owner                = "scottylabs-labrador" -> null
-           repository           = "ScottyStack" -> null
        }

        # (1 unchanged block hidden)
    }

  # github_team_membership.devops_indigo0445 will be created
+ resource "github_team_membership" "devops_indigo0445" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "14706218"
+       username = "indigo0445"
    }

Plan: 2 to add, 1 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 1 to change, 0 to destroy.


Plan Summary

6 projects, 5 with changes, 1 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.35s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_devops_ch0_indigo0445 will be created + resource "null_resource" "slack_devops_ch0_indigo0445" { + id = (known after apply) + triggers = { + "channel" = "C08K3Q77ZQF" + "user" = "U09C4RTR36F" } } # null_resource.slack_devops_ch1_indigo0445 will be created + resource "null_resource" "slack_devops_ch1_indigo0445" { + id = (known after apply) + triggers = { + "channel" = "C0BNV84NUTC" + "user" = "U09C4RTR36F" } } # null_resource.slack_hub_indigo0445 will be created + resource "null_resource" "slack_hub_indigo0445" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U09C4RTR36F" } } Plan: 3 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 3 to add, 0 to change, 0 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.32s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/226/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/226/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/226/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 7.32s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_indigo0445 will be created + resource "google_cloud_identity_group_membership" "tech_indigo0445" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "rjwu@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.26s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_devops will be updated in-place ! resource "keycloak_group_memberships" "project_devops" { id = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7" ! members = [ + "rjwu", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_changelog will be updated in-place ! resource "keycloak_group_memberships" "project_devops_changelog" { id = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77" ! members = [ + "rjwu", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_documentation will be updated in-place ! resource "keycloak_group_memberships" "project_devops_documentation" { id = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f" ! members = [ + "rjwu", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_link-shortener will be updated in-place ! resource "keycloak_group_memberships" "project_devops_link-shortener" { id = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935" ! members = [ + "rjwu", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "rjwu", # (37 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 5 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 5 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.30s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.indigo0445 will be created + resource "discord_member_roles" "indigo0445" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "629857326232371210" + role { + has_role = true + role_id = "1488787087598813215" } + role { + has_role = true + role_id = "1488787088118644848" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.30s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # forgejo_team_member.devops_indigo0445 will be created + resource "forgejo_team_member" "devops_indigo0445" { + team_id = 4 + user = "rjwu" } # github_repository.slai_cmugpt_surface will be updated in-place ! resource "github_repository" "slai_cmugpt_surface" { id = "cmugpt-surface" name = "cmugpt-surface" # (39 unchanged attributes hidden) - template { - include_all_branches = false -> null - owner = "scottylabs-labrador" -> null - repository = "ScottyStack" -> null } # (1 unchanged block hidden) } # github_team_membership.devops_indigo0445 will be created + resource "github_team_membership" "devops_indigo0445" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "14706218" + username = "indigo0445" } Plan: 2 to add, 1 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 1 to change, 0 to destroy. --- ### Plan Summary 6 projects, 5 with changes, 1 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis plan

atlantis plan
Member

atlantis apply

atlantis apply

Apply Error

cannot run "apply": pull request 226 is currently locked by "plan" for commit a993bc9.
Wait until the previous command is complete and try again
**Apply Error** ``` cannot run "apply": pull request 226 is currently locked by "plan" for commit a993bc9. Wait until the previous command is complete and try again ``` - Commit: [a993bc9cce24fb70267cd0c45e599cee011e31a9](https://git.cmu.dev/ScottyLabs/governance/pulls/226/commits/a993bc9cce24fb70267cd0c45e599cee011e31a9) - Blocking job: [View Atlantis execution](https://atlantis.scottylabs.org/jobs/c22b851c-287d-496f-9ac8-afd9a787d897)

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.20s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_devops_ch0_indigo0445 will be created
+ resource "null_resource" "slack_devops_ch0_indigo0445" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08K3Q77ZQF"
          + "user"    = "U09C4RTR36F"
        }
    }

  # null_resource.slack_devops_ch1_indigo0445 will be created
+ resource "null_resource" "slack_devops_ch1_indigo0445" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0BNV84NUTC"
          + "user"    = "U09C4RTR36F"
        }
    }

  # null_resource.slack_hub_indigo0445 will be created
+ resource "null_resource" "slack_hub_indigo0445" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U09C4RTR36F"
        }
    }

Plan: 3 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 3 to add, 0 to change, 0 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.21s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.16s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_indigo0445 will be created
+ resource "google_cloud_identity_group_membership" "tech_indigo0445" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "rjwu@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.23s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_devops will be updated in-place
! resource "keycloak_group_memberships" "project_devops" {
        id       = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7"
!       members  = [
+           "rjwu",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_changelog will be updated in-place
! resource "keycloak_group_memberships" "project_devops_changelog" {
        id       = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77"
!       members  = [
+           "rjwu",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_documentation will be updated in-place
! resource "keycloak_group_memberships" "project_devops_documentation" {
        id       = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f"
!       members  = [
+           "rjwu",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_link-shortener will be updated in-place
! resource "keycloak_group_memberships" "project_devops_link-shortener" {
        id       = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935"
!       members  = [
+           "rjwu",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "rjwu",
            # (37 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 5 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 5 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.indigo0445 will be created
+ resource "discord_member_roles" "indigo0445" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "629857326232371210"

+       role {
+           has_role = true
+           role_id  = "1488787087598813215"
        }
+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.23s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # forgejo_team_member.devops_indigo0445 will be created
+ resource "forgejo_team_member" "devops_indigo0445" {
+       team_id = 4
+       user    = "rjwu"
    }

  # github_repository.slai_cmugpt_surface will be updated in-place
! resource "github_repository" "slai_cmugpt_surface" {
        id                                      = "cmugpt-surface"
        name                                    = "cmugpt-surface"
        # (39 unchanged attributes hidden)

-       template {
-           include_all_branches = false -> null
-           owner                = "scottylabs-labrador" -> null
-           repository           = "ScottyStack" -> null
        }

        # (1 unchanged block hidden)
    }

  # github_team_membership.devops_indigo0445 will be created
+ resource "github_team_membership" "devops_indigo0445" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "14706218"
+       username = "indigo0445"
    }

Plan: 2 to add, 1 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 1 to change, 0 to destroy.


Plan Summary

6 projects, 5 with changes, 1 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.20s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_devops_ch0_indigo0445 will be created + resource "null_resource" "slack_devops_ch0_indigo0445" { + id = (known after apply) + triggers = { + "channel" = "C08K3Q77ZQF" + "user" = "U09C4RTR36F" } } # null_resource.slack_devops_ch1_indigo0445 will be created + resource "null_resource" "slack_devops_ch1_indigo0445" { + id = (known after apply) + triggers = { + "channel" = "C0BNV84NUTC" + "user" = "U09C4RTR36F" } } # null_resource.slack_hub_indigo0445 will be created + resource "null_resource" "slack_hub_indigo0445" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U09C4RTR36F" } } Plan: 3 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 3 to add, 0 to change, 0 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.21s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.16s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_indigo0445 will be created + resource "google_cloud_identity_group_membership" "tech_indigo0445" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "rjwu@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.23s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_devops will be updated in-place ! resource "keycloak_group_memberships" "project_devops" { id = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7" ! members = [ + "rjwu", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_changelog will be updated in-place ! resource "keycloak_group_memberships" "project_devops_changelog" { id = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77" ! members = [ + "rjwu", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_documentation will be updated in-place ! resource "keycloak_group_memberships" "project_devops_documentation" { id = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f" ! members = [ + "rjwu", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_link-shortener will be updated in-place ! resource "keycloak_group_memberships" "project_devops_link-shortener" { id = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935" ! members = [ + "rjwu", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "rjwu", # (37 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 5 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 5 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.indigo0445 will be created + resource "discord_member_roles" "indigo0445" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "629857326232371210" + role { + has_role = true + role_id = "1488787087598813215" } + role { + has_role = true + role_id = "1488787088118644848" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.23s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # forgejo_team_member.devops_indigo0445 will be created + resource "forgejo_team_member" "devops_indigo0445" { + team_id = 4 + user = "rjwu" } # github_repository.slai_cmugpt_surface will be updated in-place ! resource "github_repository" "slai_cmugpt_surface" { id = "cmugpt-surface" name = "cmugpt-surface" # (39 unchanged attributes hidden) - template { - include_all_branches = false -> null - owner = "scottylabs-labrador" -> null - repository = "ScottyStack" -> null } # (1 unchanged block hidden) } # github_team_membership.devops_indigo0445 will be created + resource "github_team_membership" "devops_indigo0445" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "14706218" + username = "indigo0445" } Plan: 2 to add, 1 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 1 to change, 0 to destroy. --- ### Plan Summary 6 projects, 5 with changes, 1 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis apply

atlantis apply

Ran Apply for 6 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.16s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

2. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.20s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_devops_ch1_indigo0445: Creating...
null_resource.slack_hub_indigo0445: Creating...
null_resource.slack_devops_ch0_indigo0445: Creating...
null_resource.slack_devops_ch0_indigo0445: Provisioning with 'local-exec'...
null_resource.slack_devops_ch0_indigo0445 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08K3Q77ZQF --user U09C4RTR36F"]
null_resource.slack_hub_indigo0445: Provisioning with 'local-exec'...
null_resource.slack_hub_indigo0445 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U09C4RTR36F"]
null_resource.slack_devops_ch1_indigo0445: Provisioning with 'local-exec'...
null_resource.slack_devops_ch1_indigo0445 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0BNV84NUTC --user U09C4RTR36F"]
null_resource.slack_devops_ch1_indigo0445: Creation complete after 1s [id=4030560679133036267]
null_resource.slack_hub_indigo0445: Creation complete after 1s [id=7973140652129308373]
null_resource.slack_devops_ch0_indigo0445: Creation complete after 1s [id=6369468938332311989]

Apply complete! Resources: 3 added, 0 changed, 0 destroyed.

3. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.26s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.indigo0445: Creating...
discord_member_roles.indigo0445: Creation complete after 1s [id=1353118209024331948:629857326232371210]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.14s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_indigo0445: Creating...
google_cloud_identity_group_membership.tech_indigo0445: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/100660819832913404095]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

5. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.31s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_devops_changelog: Modifying... [id=scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77]
keycloak_group_memberships.project_devops_documentation: Modifying... [id=scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f]
keycloak_group_memberships.project_devops: Modifying... [id=scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7]
keycloak_group_memberships.project_devops_link-shortener: Modifying... [id=scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_devops_documentation: Modifications complete after 2s [id=scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f]
keycloak_group_memberships.project_devops: Modifications complete after 2s [id=scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7]
keycloak_group_memberships.project_devops_changelog: Modifications complete after 2s [id=scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77]
keycloak_group_memberships.project_devops_link-shortener: Modifications complete after 2s [id=scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935]
keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 5 changed, 0 destroyed.

6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.22s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
github_repository.slai_cmugpt_surface: Modifying... [id=cmugpt-surface]
github_team_membership.devops_indigo0445: Creating...
forgejo_team_member.devops_indigo0445: Creating...
forgejo_team_member.devops_indigo0445: Creation complete after 0s
github_repository.slai_cmugpt_surface: Modifications complete after 3s [id=cmugpt-surface]
github_team_membership.devops_indigo0445: Creation complete after 3s [id=14706218:indigo0445]

Apply complete! Resources: 2 added, 1 changed, 0 destroyed.

Apply Summary

6 projects, 6 successful, 0 failed, 0 errored

Ran Apply for 6 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.16s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.20s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_devops_ch1_indigo0445: Creating... null_resource.slack_hub_indigo0445: Creating... null_resource.slack_devops_ch0_indigo0445: Creating... null_resource.slack_devops_ch0_indigo0445: Provisioning with 'local-exec'... null_resource.slack_devops_ch0_indigo0445 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08K3Q77ZQF --user U09C4RTR36F"] null_resource.slack_hub_indigo0445: Provisioning with 'local-exec'... null_resource.slack_hub_indigo0445 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U09C4RTR36F"] null_resource.slack_devops_ch1_indigo0445: Provisioning with 'local-exec'... null_resource.slack_devops_ch1_indigo0445 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0BNV84NUTC --user U09C4RTR36F"] null_resource.slack_devops_ch1_indigo0445: Creation complete after 1s [id=4030560679133036267] null_resource.slack_hub_indigo0445: Creation complete after 1s [id=7973140652129308373] null_resource.slack_devops_ch0_indigo0445: Creation complete after 1s [id=6369468938332311989] Apply complete! Resources: 3 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.26s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.indigo0445: Creating... discord_member_roles.indigo0445: Creation complete after 1s [id=1353118209024331948:629857326232371210] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.14s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_indigo0445: Creating... google_cloud_identity_group_membership.tech_indigo0445: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/100660819832913404095] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.31s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_devops_changelog: Modifying... [id=scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77] keycloak_group_memberships.project_devops_documentation: Modifying... [id=scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f] keycloak_group_memberships.project_devops: Modifying... [id=scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7] keycloak_group_memberships.project_devops_link-shortener: Modifying... [id=scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_devops_documentation: Modifications complete after 2s [id=scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f] keycloak_group_memberships.project_devops: Modifications complete after 2s [id=scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7] keycloak_group_memberships.project_devops_changelog: Modifications complete after 2s [id=scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77] keycloak_group_memberships.project_devops_link-shortener: Modifications complete after 2s [id=scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935] keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 5 changed, 0 destroyed. ``` </details> --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.22s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ github_repository.slai_cmugpt_surface: Modifying... [id=cmugpt-surface] github_team_membership.devops_indigo0445: Creating... forgejo_team_member.devops_indigo0445: Creating... forgejo_team_member.devops_indigo0445: Creation complete after 0s github_repository.slai_cmugpt_surface: Modifications complete after 3s [id=cmugpt-surface] github_team_membership.devops_indigo0445: Creation complete after 3s [id=14706218:indigo0445] Apply complete! Resources: 2 added, 1 changed, 0 destroyed. ``` </details> --- ### Apply Summary 6 projects, 6 successful, 0 failed, 0 errored

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
brightz referenced this pull request from a commit 2026-09-26 22:03:28 +00:00
Sign in to join this conversation.
No description provided.