chore: update lost and found repo #265
No reviewers
Labels
No labels
needs/linking
needs/docs
needs/research
needs/upstream
type/bug
type/external
type/feature
type/refactor
urgency
high
urgency
immediate
urgency
low
urgency
medium
No milestone
No project
No assignees
2 participants
Notifications
Due date
No due date set.
Dependencies
No dependencies set
Reference
ScottyLabs/governance!265
Loading…
Reference in a new issue
No description provided.
Delete branch "lost-and-found"
Deleting a branch is permanent. Although the deleted branch may continue to exist for a short time before it actually gets removed, it CANNOT be undone in most cases. Continue?
Ran Plan for 6 projects:
discorddir:tofu/discordworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultgitdir:tofu/gitworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultslackdir:tofu/slackworkspace:defaultgoogle_playdir:tofu/google_playworkspace:default1. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #226. To continue, delete the lock from #226 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #226. To continue, delete the lock from #226 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #226. To continue, delete the lock from #226 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #226. To continue, delete the lock from #226 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #226. To continue, delete the lock from #226 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.6. project:
google_playdir:tofu/google_playworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #226. To continue, delete the lock from #226 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.Plan Summary
6 projects, 0 with changes, 0 with no changes, 6 failed
7c6672a37f42a615e65542a615e65591d749937891d7499378a7bdac1458a7bdac1458f91454c9dbRan Plan for 9 projects:
slackdir:tofu/slackworkspace:defaultdiscorddir:tofu/discordworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:defaultopenbaodir:tofu/openbaoworkspace:defaultcdndir:tofu/cdnworkspace:defaultposthogdir:tofu/posthogworkspace:defaultmatrix_bridgesdir:tofu/matrix_bridgesworkspace:defaultsentrydir:tofu/sentryworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #253. To continue, delete the lock from #253 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #253. To continue, delete the lock from #253 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #253. To continue, delete the lock from #253 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #253. To continue, delete the lock from #253 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
openbaodir:tofu/openbaoworkspace:defaultShow Output
Plan: 6 to add, 0 to change, 6 to destroy.
6. project:
cdndir:tofu/cdnworkspace:defaultShow Output
Plan: 0 to add, 4 to change, 23 to destroy.
7. project:
posthogdir:tofu/posthogworkspace:defaultShow Output
Plan: 3 to add, 0 to change, 0 to destroy.
8. project:
matrix_bridgesdir:tofu/matrix_bridgesworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 2 to destroy.
9. project:
sentrydir:tofu/sentryworkspace:defaultShow Output
Plan: 4 to add, 0 to change, 0 to destroy.
Plan Summary
9 projects, 5 with changes, 0 with no changes, 4 failed
atlantis plan
Ran Plan for 9 projects:
posthogdir:tofu/posthogworkspace:defaultmatrix_bridgesdir:tofu/matrix_bridgesworkspace:defaultopenbaodir:tofu/openbaoworkspace:defaultcdndir:tofu/cdnworkspace:defaultsentrydir:tofu/sentryworkspace:defaultslackdir:tofu/slackworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
posthogdir:tofu/posthogworkspace:defaultShow Output
Plan: 3 to add, 0 to change, 0 to destroy.
2. project:
matrix_bridgesdir:tofu/matrix_bridgesworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 2 to destroy.
3. project:
openbaodir:tofu/openbaoworkspace:defaultShow Output
Plan: 6 to add, 0 to change, 6 to destroy.
4. project:
cdndir:tofu/cdnworkspace:defaultShow Output
Plan: 0 to add, 4 to change, 23 to destroy.
5. project:
sentrydir:tofu/sentryworkspace:defaultShow Output
Plan: 4 to add, 0 to change, 0 to destroy.
6. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 1 to destroy.
7. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 38 to add, 0 to change, 38 to destroy.
8. project:
discorddir:tofu/discordworkspace:defaultShow Output
Plan: 2 to add, 1 to change, 2 to destroy.
9. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 15 to add, 0 to change, 14 to destroy.
Plan Summary
9 projects, 9 with changes, 0 with no changes, 0 failed
atlantis apply
Ran Apply for 9 projects:
slackdir:tofu/slackworkspace:defaultcdndir:tofu/cdnworkspace:defaultopenbaodir:tofu/openbaoworkspace:defaultposthogdir:tofu/posthogworkspace:defaultdiscorddir:tofu/discordworkspace:defaultmatrix_bridgesdir:tofu/matrix_bridgesworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultsentrydir:tofu/sentryworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
2. project:
cdndir:tofu/cdnworkspace:defaultShow Output
3. project:
openbaodir:tofu/openbaoworkspace:defaultShow Output
4. project:
posthogdir:tofu/posthogworkspace:defaultShow Output
5. project:
discorddir:tofu/discordworkspace:defaultShow Output
6. project:
matrix_bridgesdir:tofu/matrix_bridgesworkspace:defaultShow Output
7. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
8. project:
sentrydir:tofu/sentryworkspace:defaultShow Output
9. project:
gitdir:tofu/gitworkspace:defaultApply Error
Show Output
Apply Summary
9 projects, 8 successful, 0 failed, 1 errored
atlantis unlock
All Atlantis locks for this PR have been unlocked and plans discarded
76a61040fc8fd3aff6b5Ran Plan for 9 projects:
posthogdir:tofu/posthogworkspace:defaultopenbaodir:tofu/openbaoworkspace:defaultcdndir:tofu/cdnworkspace:defaultmatrix_bridgesdir:tofu/matrix_bridgesworkspace:defaultsentrydir:tofu/sentryworkspace:defaultslackdir:tofu/slackworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
posthogdir:tofu/posthogworkspace:default2. project:
openbaodir:tofu/openbaoworkspace:default3. project:
cdndir:tofu/cdnworkspace:defaultShow Output
Plan: 0 to add, 4 to change, 0 to destroy.
4. project:
matrix_bridgesdir:tofu/matrix_bridgesworkspace:default5. project:
sentrydir:tofu/sentryworkspace:default6. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 1 to destroy.
7. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 38 to add, 0 to change, 38 to destroy.
8. project:
discorddir:tofu/discordworkspace:defaultShow Output
Plan: 2 to add, 1 to change, 2 to destroy.
9. project:
gitdir:tofu/gitworkspace:defaultPlan Error
Show Output
Plan Summary
9 projects, 4 with changes, 4 with no changes, 1 failed
atlantis import -p git -- forgejo_repository.lnf_lost_and_found ScottyLabs/lost-and-found
atlantis import -p git forgejo_repository.lnf_lost_and_found ScottyLabs/lost-and-found
Ran Import for project:
gitdir:tofu/gitworkspace:defaultShow Output
atlantis plan
Ran Plan for 9 projects:
openbaodir:tofu/openbaoworkspace:defaultposthogdir:tofu/posthogworkspace:defaultmatrix_bridgesdir:tofu/matrix_bridgesworkspace:defaultcdndir:tofu/cdnworkspace:defaultsentrydir:tofu/sentryworkspace:defaultslackdir:tofu/slackworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:default1. project:
openbaodir:tofu/openbaoworkspace:default2. project:
posthogdir:tofu/posthogworkspace:default3. project:
matrix_bridgesdir:tofu/matrix_bridgesworkspace:default4. project:
cdndir:tofu/cdnworkspace:defaultShow Output
Plan: 0 to add, 4 to change, 0 to destroy.
5. project:
sentrydir:tofu/sentryworkspace:default6. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 1 to destroy.
7. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 38 to add, 0 to change, 38 to destroy.
8. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 12 to add, 1 to change, 2 to destroy.
9. project:
discorddir:tofu/discordworkspace:defaultShow Output
Plan: 2 to add, 1 to change, 2 to destroy.
Plan Summary
9 projects, 5 with changes, 4 with no changes, 0 failed
atlantis apply
Ran Apply for 9 projects:
sentrydir:tofu/sentryworkspace:defaultopenbaodir:tofu/openbaoworkspace:defaultmatrix_bridgesdir:tofu/matrix_bridgesworkspace:defaultposthogdir:tofu/posthogworkspace:defaultcdndir:tofu/cdnworkspace:defaultslackdir:tofu/slackworkspace:defaultdiscorddir:tofu/discordworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
sentrydir:tofu/sentryworkspace:defaultShow Output
2. project:
openbaodir:tofu/openbaoworkspace:defaultShow Output
3. project:
matrix_bridgesdir:tofu/matrix_bridgesworkspace:defaultShow Output
4. project:
posthogdir:tofu/posthogworkspace:defaultShow Output
5. project:
cdndir:tofu/cdnworkspace:defaultShow Output
6. project:
slackdir:tofu/slackworkspace:defaultShow Output
7. project:
discorddir:tofu/discordworkspace:defaultShow Output
8. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
9. project:
gitdir:tofu/gitworkspace:defaultShow Output
Apply Summary
9 projects, 9 successful, 0 failed, 0 errored
Locks and plans deleted for the projects and workspaces modified in this pull request:
tofu/cdnworkspace:defaulttofu/discordworkspace:defaulttofu/gitworkspace:defaulttofu/keycloakworkspace:defaulttofu/matrix_bridgesworkspace:defaulttofu/openbaoworkspace:defaulttofu/posthogworkspace:defaulttofu/sentryworkspace:defaulttofu/slackworkspace:default