feat: add alexac to cmu-maps #433

Open
thesuperRL wants to merge 3 commits from sync/add-alexac-cmu-maps into main
Owner

Clean re-submission superseding #401. Data-only change from main; tofu regen will be committed by CI.

Clean re-submission superseding #401. Data-only change from main; tofu regen will be committed by CI.
feat: add alexac to cmu-maps
Some checks failed
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan 0/0 projects planned successfully.
atlantis/apply 0/0 projects applied successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
CI / check (pull_request) Has been cancelled
kennel/build skipped: no services or static sites defined
CI / check-1 (pull_request) Has been cancelled
Validate / Governance (pull_request_target) Has been cancelled
dfb60875f1
Author
Owner

atlantis plan

atlantis plan

Ran Plan for 0 projects:

Ran Plan for 0 projects:
fix: correct TOML for alexac addition
All checks were successful
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
Validate / Governance (pull_request_target) Successful in 2m59s
CI / check-1 (pull_request) Successful in 4m7s
CI / check (pull_request) Successful in 0s
CI / build (pull_request) Has been skipped
atlantis/plan 0/0 projects planned successfully.
atlantis/apply 0/0 projects applied successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
kennel/build skipped: no services or static sites defined
6867e71a60
Author
Owner

atlantis plan

atlantis plan

Ran Plan for 0 projects:

Ran Plan for 0 projects:
Author
Owner

atlantis unlock

atlantis unlock

All Atlantis locks for this PR have been unlocked and plans discarded

All Atlantis locks for this PR have been unlocked and plans discarded
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
atlantis/plan: discord Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: google_play Plan: 2 to add, 0 to change, 2 to destroy.
atlantis/plan 5/6 projects failed to plan.
atlantis/plan: slack Plan failed.
atlantis/plan: git Plan failed.
kennel/build skipped: no services or static sites defined
atlantis/plan: keycloak Plan failed.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
fefc7ce637

Ran Plan for 6 projects:

  1. project: discord dir: tofu/discord workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: google_play dir: tofu/google_play workspace: default

1. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: google_play dir: tofu/google_play workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.07s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # googleplay_app_iam.eats_stanleymw will be destroyed
  # (because googleplay_app_iam.eats_stanleymw is not in configuration)
- resource "googleplay_app_iam" "eats_stanleymw" {
-       app_id               = "com.cmueats.app" -> null
-       expanded_permissions = [
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       permissions          = [
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       user_id              = "smwang2@andrew.cmu.edu" -> null
    }

  # googleplay_app_iam.maps_alexac will be created
+ resource "googleplay_app_iam" "maps_alexac" {
+       app_id               = "com.cmumaps.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "alexac@andrew.cmu.edu"
    }

  # googleplay_user.alexac will be created
+ resource "googleplay_user" "alexac" {
+       email                = "alexac@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

  # googleplay_user.stanleymw will be destroyed
  # (because googleplay_user.stanleymw is not in configuration)
- resource "googleplay_user" "stanleymw" {
-       email                = "smwang2@andrew.cmu.edu" -> null
-       expanded_permissions = [
-           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ] -> null
-       global_permissions   = [
-           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ] -> null
-       id                   = "smwang2@andrew.cmu.edu" -> null
-       name                 = "developers/7984360456919420393/users/smwang2@andrew.cmu.edu" -> null
    }

Plan: 2 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 2 to add, 0 to change, 2 to destroy.


Plan Summary

6 projects, 1 with changes, 0 with no changes, 5 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` --- ### 1. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.07s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # googleplay_app_iam.eats_stanleymw will be destroyed # (because googleplay_app_iam.eats_stanleymw is not in configuration) - resource "googleplay_app_iam" "eats_stanleymw" { - app_id = "com.cmueats.app" -> null - expanded_permissions = [ - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - permissions = [ - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - user_id = "smwang2@andrew.cmu.edu" -> null } # googleplay_app_iam.maps_alexac will be created + resource "googleplay_app_iam" "maps_alexac" { + app_id = "com.cmumaps.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "alexac@andrew.cmu.edu" } # googleplay_user.alexac will be created + resource "googleplay_user" "alexac" { + email = "alexac@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } # googleplay_user.stanleymw will be destroyed # (because googleplay_user.stanleymw is not in configuration) - resource "googleplay_user" "stanleymw" { - email = "smwang2@andrew.cmu.edu" -> null - expanded_permissions = [ - "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] -> null - global_permissions = [ - "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] -> null - id = "smwang2@andrew.cmu.edu" -> null - name = "developers/7984360456919420393/users/smwang2@andrew.cmu.edu" -> null } Plan: 2 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 2 to add, 0 to change, 2 to destroy. --- ### Plan Summary 6 projects, 1 with changes, 0 with no changes, 5 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis unlock

atlantis unlock

All Atlantis locks for this PR have been unlocked and plans discarded

All Atlantis locks for this PR have been unlocked and plans discarded
Some checks failed
atlantis/plan: discord Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: google_play Plan: 2 to add, 0 to change, 2 to destroy.
atlantis/plan 5/6 projects failed to plan.
atlantis/plan: slack Plan failed.
atlantis/plan: git Plan failed.
kennel/build skipped: no services or static sites defined
atlantis/plan: keycloak Plan failed.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
This pull request has changes conflicting with the target branch.
  • tofu/google_play/google_play.tf.json
  • tofu/keycloak/keycloak_memberships.tf.json
View command line instructions

Manual merge helper

Use this merge commit message when completing the merge manually.

Checkout

From your project repository, check out a new branch and test the changes.
git fetch -u origin sync/add-alexac-cmu-maps:sync/add-alexac-cmu-maps
git switch sync/add-alexac-cmu-maps

Merge

Merge the changes and update on Forgejo.

Warning: The "Autodetect manual merge" setting is not enabled for this repository, you will have to mark this pull request as manually merged afterwards.

git switch main
git merge --no-ff sync/add-alexac-cmu-maps
git switch sync/add-alexac-cmu-maps
git rebase main
git switch main
git merge --ff-only sync/add-alexac-cmu-maps
git switch sync/add-alexac-cmu-maps
git rebase main
git switch main
git merge --no-ff sync/add-alexac-cmu-maps
git switch main
git merge --squash sync/add-alexac-cmu-maps
git switch main
git merge --ff-only sync/add-alexac-cmu-maps
git switch main
git merge sync/add-alexac-cmu-maps
git push origin main
Sign in to join this conversation.
No description provided.