feat: add samtran to cbp #434
No reviewers
Labels
No labels
needs/linking
needs/docs
needs/research
needs/upstream
type/bug
type/external
type/feature
type/refactor
urgency
high
urgency
immediate
urgency
low
urgency
medium
No milestone
No project
No assignees
3 participants
Notifications
Due date
No due date set.
Dependencies
No dependencies set
Reference
ScottyLabs/governance!434
Loading…
Reference in a new issue
No description provided.
Delete branch "sync/add-samtran-cbp"
Deleting a branch is permanent. Although the deleted branch may continue to exist for a short time before it actually gets removed, it CANNOT be undone in most cases. Continue?
Clean re-submission superseding #382. Data-only change from main; tofu regen will be committed by CI.
atlantis plan
Ran Plan for 0 projects:
atlantis plan
Ran Plan for 0 projects:
atlantis unlock
All Atlantis locks for this PR have been unlocked and plans discarded
Ran Plan for 5 projects:
slackdir:tofu/slackworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 1 to destroy.
2. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 0 to add, 6 to change, 0 to destroy.
3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 1 to destroy.
4. project:
gitdir:tofu/gitworkspace:defaultPlan Error
Show Output
5. project:
discorddir:tofu/discordworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 1 to destroy.
Plan Summary
5 projects, 4 with changes, 0 with no changes, 1 failed
atlantis plan
Ran Plan for 5 projects:
slackdir:tofu/slackworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 1 to destroy.
2. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 0 to add, 6 to change, 0 to destroy.
3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 1 to destroy.
4. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 1 to destroy.
5. project:
discorddir:tofu/discordworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 1 to destroy.
Plan Summary
5 projects, 5 with changes, 0 with no changes, 0 failed
Plan Error
Plan Error
atlantis unlock
atlantis plan
Ran Plan for 5 projects:
slackdir:tofu/slackworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 1 to destroy.
2. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Error
Show Output
3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Error
Show Output
4. project:
gitdir:tofu/gitworkspace:defaultPlan Error
Show Output
5. project:
discorddir:tofu/discordworkspace:defaultPlan Error
Show Output
Plan Summary
5 projects, 1 with changes, 0 with no changes, 4 failed
All Atlantis locks for this PR have been unlocked and plans discarded
atlantis unlock
All Atlantis locks for this PR have been unlocked and plans discarded
atlantis plan
Ran Plan for 5 projects:
slackdir:tofu/slackworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultPlan Error
Show Output
2. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Error
Show Output
3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Error
Show Output
4. project:
gitdir:tofu/gitworkspace:defaultPlan Error
Show Output
5. project:
discorddir:tofu/discordworkspace:defaultPlan Error
Show Output
Plan Summary
5 projects, 0 with changes, 0 with no changes, 5 failed
Plan Error
Plan Error
atlantis unlock
All Atlantis locks for this PR have been unlocked and plans discarded
atlantis plan
Ran Plan for 5 projects:
discorddir:tofu/discordworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultgitdir:tofu/gitworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultslackdir:tofu/slackworkspace:default1. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #420. To continue, delete the lock from #420 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #420. To continue, delete the lock from #420 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #420. To continue, delete the lock from #420 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #420. To continue, delete the lock from #420 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #420. To continue, delete the lock from #420 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.Plan Summary
5 projects, 0 with changes, 0 with no changes, 5 failed
atlantis unlock
All Atlantis locks for this PR have been unlocked and plans discarded
atlantis plan
Ran Plan for 5 projects:
slackdir:tofu/slackworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 3 to destroy.
2. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 0 to add, 7 to change, 0 to destroy.
3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 2 to destroy.
4. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 14 to add, 1 to change, 2 to destroy.
5. project:
discorddir:tofu/discordworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 1 to destroy.
Plan Summary
5 projects, 5 with changes, 0 with no changes, 0 failed
Plan Error
atlantis unlock
All Atlantis locks for this PR have been unlocked and plans discarded
atlantis plan
Plan Error
Ran Plan for 5 projects:
slackdir:tofu/slackworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultPlan Error
Show Output
2. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Error
Show Output
3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Error
Show Output
4. project:
gitdir:tofu/gitworkspace:defaultPlan Error
Show Output
5. project:
discorddir:tofu/discordworkspace:defaultPlan Error
Show Output
Plan Summary
5 projects, 0 with changes, 0 with no changes, 5 failed
atlantis unlock
All Atlantis locks for this PR have been unlocked and plans discarded
atlantis plan
Plan Error
Ran Plan for 5 projects:
slackdir:tofu/slackworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 0 to destroy.
2. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 0 to destroy.
3. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 0 to add, 5 to change, 0 to destroy.
4. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 3 to add, 1 to change, 0 to destroy.
5. project:
discorddir:tofu/discordworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 0 to destroy.
Plan Summary
5 projects, 5 with changes, 0 with no changes, 0 failed
atlantis apply
Ran Apply for 5 projects:
keycloakdir:tofu/keycloakworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgitdir:tofu/gitworkspace:defaultslackdir:tofu/slackworkspace:default1. project:
keycloakdir:tofu/keycloakworkspace:defaultApply Failed: Pull request must be mergeable before running apply.
2. project:
google_groupsdir:tofu/google_groupsworkspace:defaultApply Failed: Pull request must be mergeable before running apply.
3. project:
discorddir:tofu/discordworkspace:defaultApply Failed: Pull request must be mergeable before running apply.
4. project:
gitdir:tofu/gitworkspace:defaultApply Failed: Pull request must be mergeable before running apply.
5. project:
slackdir:tofu/slackworkspace:defaultApply Failed: Pull request must be mergeable before running apply.
Apply Summary
5 projects, 0 successful, 5 failed, 0 errored
Ran Plan for 5 projects:
keycloakdir:tofu/keycloakworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultslackdir:tofu/slackworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.Plan Summary
5 projects, 0 with changes, 0 with no changes, 5 failed
atlantis unlock
All Atlantis locks for this PR have been unlocked and plans discarded
atlantis plan
Ran Plan for 5 projects:
slackdir:tofu/slackworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.Plan Summary
5 projects, 0 with changes, 0 with no changes, 5 failed
atlantis unlock
All Atlantis locks for this PR have been unlocked and plans discarded
atlantis plan
Ran Plan for 5 projects:
discorddir:tofu/discordworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultslackdir:tofu/slackworkspace:default1. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #448. To continue, delete the lock from #448 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.Plan Summary
5 projects, 0 with changes, 0 with no changes, 5 failed
atlantis plan
Ran Plan for 5 projects:
slackdir:tofu/slackworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 3 to add, 0 to change, 2 to destroy.
2. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 1 to destroy.
3. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 0 to add, 6 to change, 0 to destroy.
4. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 4 to add, 1 to change, 2 to destroy.
5. project:
discorddir:tofu/discordworkspace:defaultShow Output
Plan: 2 to add, 1 to change, 1 to destroy.
Plan Summary
5 projects, 5 with changes, 0 with no changes, 0 failed
Ran Plan for 6 projects:
google_playdir:tofu/google_playworkspace:defaultslackdir:tofu/slackworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:default1. project:
google_playdir:tofu/google_playworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #433. To continue, delete the lock from #433 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 3 to add, 0 to change, 2 to destroy.
3. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 0 to add, 6 to change, 0 to destroy.
4. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 1 to destroy.
5. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 2 to destroy.
6. project:
discorddir:tofu/discordworkspace:defaultShow Output
Plan: 2 to add, 1 to change, 1 to destroy.
Plan Summary
6 projects, 5 with changes, 0 with no changes, 1 failed
atlantis unlock
All Atlantis locks for this PR have been unlocked and plans discarded
0f20e965e783f6130867Ran Plan for 6 projects:
keycloakdir:tofu/keycloakworkspace:defaultslackdir:tofu/slackworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgitdir:tofu/gitworkspace:defaultgoogle_playdir:tofu/google_playworkspace:default1. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #450. To continue, delete the lock from #450 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #450. To continue, delete the lock from #450 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #450. To continue, delete the lock from #450 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #450. To continue, delete the lock from #450 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #450. To continue, delete the lock from #450 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.6. project:
google_playdir:tofu/google_playworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 0 to destroy.
Plan Summary
6 projects, 1 with changes, 0 with no changes, 5 failed
atlantis apply
Apply Error
atlantis plan
Ran Plan for 6 projects:
slackdir:tofu/slackworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgoogle_playdir:tofu/google_playworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 4 to add, 0 to change, 0 to destroy.
2. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 0 to add, 6 to change, 0 to destroy.
3. project:
google_playdir:tofu/google_playworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 0 to destroy.
4. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 0 to destroy.
5. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 0 to destroy.
6. project:
discorddir:tofu/discordworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 0 to destroy.
Plan Summary
6 projects, 6 with changes, 0 with no changes, 0 failed
atlantis apply
Ran Apply for 6 projects:
slackdir:tofu/slackworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgoogle_playdir:tofu/google_playworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultkeycloakdir:tofu/keycloakworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
2. project:
gitdir:tofu/gitworkspace:defaultShow Output
3. project:
discorddir:tofu/discordworkspace:defaultApply Error
Show Output
4. project:
google_playdir:tofu/google_playworkspace:defaultShow Output
5. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
6. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Apply Summary
6 projects, 5 successful, 0 failed, 1 errored
Locks and plans deleted for the projects and workspaces modified in this pull request:
tofu/discordworkspace:defaulttofu/gitworkspace:defaulttofu/google_groupsworkspace:defaulttofu/google_playworkspace:defaulttofu/keycloakworkspace:defaulttofu/slackworkspace:default