feat: add stanleymw to cmueats #420

Merged
thesuperRL merged 2 commits from stanleymw/governance:main into main 2026-09-27 19:23:44 +00:00
Contributor
No description provided.
feat: add stanleymw to cmueats
Some checks are pending
CI / check (pull_request) Blocked by required conditions
CI / check-1 (pull_request) Blocked by required conditions
CI / build (pull_request) Blocked by required conditions
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 3m8s
36801cac9f
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/apply Apply failed.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: slack Plan: 2 to add, 0 to change, 4 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 8 to change, 0 to destroy.
atlantis/plan: google_play Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 2 to destroy.
atlantis/plan: git Plan: 1 to add, 0 to change, 2 to destroy.
atlantis/plan: discord Plan: 1 to add, 1 to change, 2 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
abb548b0cf

Ran Plan for 6 projects:

  1. project: google_groups dir: tofu/google_groups workspace: default
  2. project: git dir: tofu/git workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: slack dir: tofu/slack workspace: default
  6. project: google_play dir: tofu/google_play workspace: default

1. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #390. To continue, delete the lock from #390 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #390. To continue, delete the lock from #390 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #390. To continue, delete the lock from #390 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #390. To continue, delete the lock from #390 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #390. To continue, delete the lock from #390 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` --- ### 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #390. To continue, delete the lock from #390 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #390. To continue, delete the lock from #390 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #390. To continue, delete the lock from #390 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #390. To continue, delete the lock from #390 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #390. To continue, delete the lock from #390 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis apply

atlantis apply

Apply Error

plan for dir "tofu/google_groups" workspace "default" project "google_groups" errored and cannot be applied without a plan file; run `atlantis plan`
**Apply Error** ``` plan for dir "tofu/google_groups" workspace "default" project "google_groups" errored and cannot be applied without a plan file; run `atlantis plan` ```
Member

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: git dir: tofu/git workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: slack dir: tofu/slack workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: keycloak dir: tofu/keycloak workspace: default

1. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #423. To continue, delete the lock from #423 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #417. To continue, delete the lock from #417 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #423. To continue, delete the lock from #423 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #423. To continue, delete the lock from #423 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #423. To continue, delete the lock from #423 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #423. To continue, delete the lock from #423 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #417. To continue, delete the lock from #417 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #423. To continue, delete the lock from #423 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #423. To continue, delete the lock from #423 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #423. To continue, delete the lock from #423 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis apply

atlantis apply

Apply Error

plan for dir "tofu/google_groups" workspace "default" project "google_groups" errored and cannot be applied without a plan file; run `atlantis plan`
**Apply Error** ``` plan for dir "tofu/google_groups" workspace "default" project "google_groups" errored and cannot be applied without a plan file; run `atlantis plan` ```
Owner

atlantis plan

atlantis plan
Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/420/default": 
Download attempt 1 failed, retrying...
Download attempt 2 failed, retrying...
Download attempt 3 failed, retrying...
Failed to download forgejo provider after 3 attempts

``` Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/420/default": Download attempt 1 failed, retrying... Download attempt 2 failed, retrying... Download attempt 3 failed, retrying... Failed to download forgejo provider after 3 attempts ```
Owner

atlantis plan

atlantis plan
Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/420/default": 
Download attempt 1 failed, retrying...
Download attempt 2 failed, retrying...
Download attempt 3 failed, retrying...
Failed to download forgejo provider after 3 attempts

``` Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/420/default": Download attempt 1 failed, retrying... Download attempt 2 failed, retrying... Download attempt 3 failed, retrying... Failed to download forgejo provider after 3 attempts ```
Owner

atlantis plan

atlantis plan
Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/420/default": 
Download attempt 1 failed, retrying...
Download attempt 2 failed, retrying...
Download attempt 3 failed, retrying...
Failed to download forgejo provider after 3 attempts

``` Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/420/default": Download attempt 1 failed, retrying... Download attempt 2 failed, retrying... Download attempt 3 failed, retrying... Failed to download forgejo provider after 3 attempts ```
Owner

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: git dir: tofu/git workspace: default
  3. project: keycloak dir: tofu/keycloak workspace: default
  4. project: slack dir: tofu/slack workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: google_groups dir: tofu/google_groups workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #262. To continue, delete the lock from #262 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #262. To continue, delete the lock from #262 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: keycloak dir: tofu/keycloak workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: discord dir: tofu/discord workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #380. To continue, delete the lock from #380 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.19s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-eats_ch0_stanleymw will be created
+ resource "null_resource" "slack_cmu-eats_ch0_stanleymw" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C043UV30DHA"
          + "user"    = "U09DUH5550B"
        }
    }

  # null_resource.slack_devops_ch0_brightz will be destroyed
  # (because null_resource.slack_devops_ch0_brightz is not in configuration)
- resource "null_resource" "slack_devops_ch0_brightz" {
-       id       = "1977460316515527708" -> null
-       triggers = {
          - "channel" = "C08K3Q77ZQF"
          - "user"    = "U09F3P78LT0"
        } -> null
    }

  # null_resource.slack_devops_ch1_brightz will be destroyed
  # (because null_resource.slack_devops_ch1_brightz is not in configuration)
- resource "null_resource" "slack_devops_ch1_brightz" {
-       id       = "4471079696359905776" -> null
-       triggers = {
          - "channel" = "C0BNV84NUTC"
          - "user"    = "U09F3P78LT0"
        } -> null
    }

  # null_resource.slack_hub_locho will be destroyed
  # (because null_resource.slack_hub_locho is not in configuration)
- resource "null_resource" "slack_hub_locho" {
-       id       = "4360225298301891996" -> null
-       triggers = {
          - "channel" = "C096TM8EMS8"
          - "user"    = "U0C12EHAGPP"
        } -> null
    }

  # null_resource.slack_hub_lwahi will be destroyed
  # (because null_resource.slack_hub_lwahi is not in configuration)
- resource "null_resource" "slack_hub_lwahi" {
-       id       = "982101409620673204" -> null
-       triggers = {
          - "channel" = "C096TM8EMS8"
          - "user"    = "U0C18F0GT7G"
        } -> null
    }

  # null_resource.slack_hub_stanleymw will be created
+ resource "null_resource" "slack_hub_stanleymw" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U09DUH5550B"
        }
    }

Plan: 2 to add, 0 to change, 4 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 4 to destroy.


3. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-eats_eats will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-eats_eats" {
        id       = "scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba"
!       members  = [
+           "smwang2",
            # (7 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-housing_housing" {
        id       = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d"
!       members  = [
-           "lxh",
            # (14 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops will be updated in-place
! resource "keycloak_group_memberships" "project_devops" {
        id       = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7"
!       members  = [
-           "brightz",
            # (10 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_changelog will be updated in-place
! resource "keycloak_group_memberships" "project_devops_changelog" {
        id       = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77"
!       members  = [
-           "brightz",
            # (10 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_documentation will be updated in-place
! resource "keycloak_group_memberships" "project_devops_documentation" {
        id       = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f"
!       members  = [
-           "brightz",
            # (10 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_link-shortener will be updated in-place
! resource "keycloak_group_memberships" "project_devops_link-shortener" {
        id       = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935"
!       members  = [
-           "brightz",
            # (10 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_tartan-vote will be updated in-place
! resource "keycloak_group_memberships" "project_tartan-vote" {
        id       = "scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c"
!       members  = [
-           "lwahi",
            # (10 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
-           "lwahi",
-           "lxh",
+           "smwang2",
            # (94 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 8 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 8 to change, 0 to destroy.


4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on shared package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_locho will be destroyed
  # (because google_cloud_identity_group_membership.tech_locho is not in configuration)
- resource "google_cloud_identity_group_membership" "tech_locho" {
-       create_time = "2026-09-27T18:14:27.052016Z" -> null
-       group       = "groups/04anzqyu1bqbv79" -> null
-       id          = "groups/04anzqyu1bqbv79/memberships/111069990805475038275" -> null
-       name        = "groups/04anzqyu1bqbv79/memberships/111069990805475038275" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-27T18:14:27.052016Z" -> null

-       preferred_member_key {
-           id = "lxh@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.tech_lwahi will be destroyed
  # (because google_cloud_identity_group_membership.tech_lwahi is not in configuration)
- resource "google_cloud_identity_group_membership" "tech_lwahi" {
-       create_time = "2026-09-26T23:42:21.456782Z" -> null
-       group       = "groups/04anzqyu1bqbv79" -> null
-       id          = "groups/04anzqyu1bqbv79/memberships/112352542527683783999" -> null
-       name        = "groups/04anzqyu1bqbv79/memberships/112352542527683783999" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-26T23:42:21.456782Z" -> null

-       preferred_member_key {
-           id = "lwahi@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.tech_stanleymw will be created
+ resource "google_cloud_identity_group_membership" "tech_stanleymw" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "smwang2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 2 to destroy.


5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.21s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # forgejo_team_member.cmu-eats_stanleymw will be created
+ resource "forgejo_team_member" "cmu-eats_stanleymw" {
+       team_id = 28
+       user    = "stanleymw"
    }

  # forgejo_team_member.cmu-housing_locho will be destroyed
  # (because forgejo_team_member.cmu-housing_locho is not in configuration)
- resource "forgejo_team_member" "cmu-housing_locho" {
-       team_id = 2 -> null
-       user    = "locho" -> null
    }

  # forgejo_team_member.devops_brightz will be destroyed
  # (because forgejo_team_member.devops_brightz is not in configuration)
- resource "forgejo_team_member" "devops_brightz" {
-       team_id = 4 -> null
-       user    = "brightz" -> null
    }

Plan: 1 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 1 to add, 0 to change, 2 to destroy.


6. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.15s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)
- destroy

OpenTofu will perform the following actions:

  # discord_member_roles.brightz will be updated in-place
! resource "discord_member_roles" "brightz" {
        id        = "1353118209024331948:689114594794668076"
        # (2 unchanged attributes hidden)

-       role {
-           has_role = true -> null
-           role_id  = "1488787087598813215" -> null
        }

        # (2 unchanged blocks hidden)
    }

  # discord_member_roles.locho will be destroyed
  # (because discord_member_roles.locho is not in configuration)
- resource "discord_member_roles" "locho" {
-       id        = "1353118209024331948:582666638550368277" -> null
-       server_id = "1353118209024331948" -> null
-       user_id   = "582666638550368277" -> null

-       role {
-           has_role = true -> null
-           role_id  = "1488787088118644848" -> null
        }
-       role {
-           has_role = true -> null
-           role_id  = "1513653499957874871" -> null
        }
    }

  # discord_member_roles.lwahi will be destroyed
  # (because discord_member_roles.lwahi is not in configuration)
- resource "discord_member_roles" "lwahi" {
-       id        = "1353118209024331948:738509196487688262" -> null
-       server_id = "1353118209024331948" -> null
-       user_id   = "738509196487688262" -> null

-       role {
-           has_role = true -> null
-           role_id  = "1488787088118644848" -> null
        }
-       role {
-           has_role = true -> null
-           role_id  = "1508601984255524885" -> null
        }
    }

  # discord_member_roles.stanleymw will be created
+ resource "discord_member_roles" "stanleymw" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "1367644731571568660"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1541552714063024148"
        }
    }

Plan: 1 to add, 1 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 1 to change, 2 to destroy.


Plan Summary

6 projects, 5 with changes, 0 with no changes, 1 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #380. To continue, delete the lock from #380 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.19s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # null_resource.slack_cmu-eats_ch0_stanleymw will be created + resource "null_resource" "slack_cmu-eats_ch0_stanleymw" { + id = (known after apply) + triggers = { + "channel" = "C043UV30DHA" + "user" = "U09DUH5550B" } } # null_resource.slack_devops_ch0_brightz will be destroyed # (because null_resource.slack_devops_ch0_brightz is not in configuration) - resource "null_resource" "slack_devops_ch0_brightz" { - id = "1977460316515527708" -> null - triggers = { - "channel" = "C08K3Q77ZQF" - "user" = "U09F3P78LT0" } -> null } # null_resource.slack_devops_ch1_brightz will be destroyed # (because null_resource.slack_devops_ch1_brightz is not in configuration) - resource "null_resource" "slack_devops_ch1_brightz" { - id = "4471079696359905776" -> null - triggers = { - "channel" = "C0BNV84NUTC" - "user" = "U09F3P78LT0" } -> null } # null_resource.slack_hub_locho will be destroyed # (because null_resource.slack_hub_locho is not in configuration) - resource "null_resource" "slack_hub_locho" { - id = "4360225298301891996" -> null - triggers = { - "channel" = "C096TM8EMS8" - "user" = "U0C12EHAGPP" } -> null } # null_resource.slack_hub_lwahi will be destroyed # (because null_resource.slack_hub_lwahi is not in configuration) - resource "null_resource" "slack_hub_lwahi" { - id = "982101409620673204" -> null - triggers = { - "channel" = "C096TM8EMS8" - "user" = "U0C18F0GT7G" } -> null } # null_resource.slack_hub_stanleymw will be created + resource "null_resource" "slack_hub_stanleymw" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U09DUH5550B" } } Plan: 2 to add, 0 to change, 4 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 4 to destroy. --- ### 3. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-eats_eats will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-eats_eats" { id = "scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba" ! members = [ + "smwang2", # (7 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-housing_housing" { id = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d" ! members = [ - "lxh", # (14 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops will be updated in-place ! resource "keycloak_group_memberships" "project_devops" { id = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7" ! members = [ - "brightz", # (10 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_changelog will be updated in-place ! resource "keycloak_group_memberships" "project_devops_changelog" { id = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77" ! members = [ - "brightz", # (10 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_documentation will be updated in-place ! resource "keycloak_group_memberships" "project_devops_documentation" { id = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f" ! members = [ - "brightz", # (10 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_link-shortener will be updated in-place ! resource "keycloak_group_memberships" "project_devops_link-shortener" { id = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935" ! members = [ - "brightz", # (10 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_tartan-vote will be updated in-place ! resource "keycloak_group_memberships" "project_tartan-vote" { id = "scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c" ! members = [ - "lwahi", # (10 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ - "lwahi", - "lxh", + "smwang2", # (94 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 8 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 8 to change, 0 to destroy. --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on shared package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_locho will be destroyed # (because google_cloud_identity_group_membership.tech_locho is not in configuration) - resource "google_cloud_identity_group_membership" "tech_locho" { - create_time = "2026-09-27T18:14:27.052016Z" -> null - group = "groups/04anzqyu1bqbv79" -> null - id = "groups/04anzqyu1bqbv79/memberships/111069990805475038275" -> null - name = "groups/04anzqyu1bqbv79/memberships/111069990805475038275" -> null - type = "USER" -> null - update_time = "2026-09-27T18:14:27.052016Z" -> null - preferred_member_key { - id = "lxh@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.tech_lwahi will be destroyed # (because google_cloud_identity_group_membership.tech_lwahi is not in configuration) - resource "google_cloud_identity_group_membership" "tech_lwahi" { - create_time = "2026-09-26T23:42:21.456782Z" -> null - group = "groups/04anzqyu1bqbv79" -> null - id = "groups/04anzqyu1bqbv79/memberships/112352542527683783999" -> null - name = "groups/04anzqyu1bqbv79/memberships/112352542527683783999" -> null - type = "USER" -> null - update_time = "2026-09-26T23:42:21.456782Z" -> null - preferred_member_key { - id = "lwahi@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.tech_stanleymw will be created + resource "google_cloud_identity_group_membership" "tech_stanleymw" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "smwang2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 2 to destroy. --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.21s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # forgejo_team_member.cmu-eats_stanleymw will be created + resource "forgejo_team_member" "cmu-eats_stanleymw" { + team_id = 28 + user = "stanleymw" } # forgejo_team_member.cmu-housing_locho will be destroyed # (because forgejo_team_member.cmu-housing_locho is not in configuration) - resource "forgejo_team_member" "cmu-housing_locho" { - team_id = 2 -> null - user = "locho" -> null } # forgejo_team_member.devops_brightz will be destroyed # (because forgejo_team_member.devops_brightz is not in configuration) - resource "forgejo_team_member" "devops_brightz" { - team_id = 4 -> null - user = "brightz" -> null } Plan: 1 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 1 to add, 0 to change, 2 to destroy. --- ### 6. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.15s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) - destroy OpenTofu will perform the following actions: # discord_member_roles.brightz will be updated in-place ! resource "discord_member_roles" "brightz" { id = "1353118209024331948:689114594794668076" # (2 unchanged attributes hidden) - role { - has_role = true -> null - role_id = "1488787087598813215" -> null } # (2 unchanged blocks hidden) } # discord_member_roles.locho will be destroyed # (because discord_member_roles.locho is not in configuration) - resource "discord_member_roles" "locho" { - id = "1353118209024331948:582666638550368277" -> null - server_id = "1353118209024331948" -> null - user_id = "582666638550368277" -> null - role { - has_role = true -> null - role_id = "1488787088118644848" -> null } - role { - has_role = true -> null - role_id = "1513653499957874871" -> null } } # discord_member_roles.lwahi will be destroyed # (because discord_member_roles.lwahi is not in configuration) - resource "discord_member_roles" "lwahi" { - id = "1353118209024331948:738509196487688262" -> null - server_id = "1353118209024331948" -> null - user_id = "738509196487688262" -> null - role { - has_role = true -> null - role_id = "1488787088118644848" -> null } - role { - has_role = true -> null - role_id = "1508601984255524885" -> null } } # discord_member_roles.stanleymw will be created + resource "discord_member_roles" "stanleymw" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "1367644731571568660" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1541552714063024148" } } Plan: 1 to add, 1 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 1 to change, 2 to destroy. --- ### Plan Summary 6 projects, 5 with changes, 0 with no changes, 1 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: keycloak dir: tofu/keycloak workspace: default
  3. project: google_play dir: tofu/google_play workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: discord dir: tofu/discord workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.17s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-eats_ch0_stanleymw will be created
+ resource "null_resource" "slack_cmu-eats_ch0_stanleymw" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C043UV30DHA"
          + "user"    = "U09DUH5550B"
        }
    }

  # null_resource.slack_devops_ch0_brightz will be destroyed
  # (because null_resource.slack_devops_ch0_brightz is not in configuration)
- resource "null_resource" "slack_devops_ch0_brightz" {
-       id       = "1977460316515527708" -> null
-       triggers = {
          - "channel" = "C08K3Q77ZQF"
          - "user"    = "U09F3P78LT0"
        } -> null
    }

  # null_resource.slack_devops_ch1_brightz will be destroyed
  # (because null_resource.slack_devops_ch1_brightz is not in configuration)
- resource "null_resource" "slack_devops_ch1_brightz" {
-       id       = "4471079696359905776" -> null
-       triggers = {
          - "channel" = "C0BNV84NUTC"
          - "user"    = "U09F3P78LT0"
        } -> null
    }

  # null_resource.slack_hub_locho will be destroyed
  # (because null_resource.slack_hub_locho is not in configuration)
- resource "null_resource" "slack_hub_locho" {
-       id       = "4360225298301891996" -> null
-       triggers = {
          - "channel" = "C096TM8EMS8"
          - "user"    = "U0C12EHAGPP"
        } -> null
    }

  # null_resource.slack_hub_lwahi will be destroyed
  # (because null_resource.slack_hub_lwahi is not in configuration)
- resource "null_resource" "slack_hub_lwahi" {
-       id       = "982101409620673204" -> null
-       triggers = {
          - "channel" = "C096TM8EMS8"
          - "user"    = "U0C18F0GT7G"
        } -> null
    }

  # null_resource.slack_hub_stanleymw will be created
+ resource "null_resource" "slack_hub_stanleymw" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U09DUH5550B"
        }
    }

Plan: 2 to add, 0 to change, 4 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 4 to destroy.


2. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.26s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-eats_eats will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-eats_eats" {
        id       = "scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba"
!       members  = [
+           "smwang2",
            # (7 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-housing_housing" {
        id       = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d"
!       members  = [
-           "lxh",
            # (14 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops will be updated in-place
! resource "keycloak_group_memberships" "project_devops" {
        id       = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7"
!       members  = [
-           "brightz",
            # (10 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_changelog will be updated in-place
! resource "keycloak_group_memberships" "project_devops_changelog" {
        id       = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77"
!       members  = [
-           "brightz",
            # (10 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_documentation will be updated in-place
! resource "keycloak_group_memberships" "project_devops_documentation" {
        id       = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f"
!       members  = [
-           "brightz",
            # (10 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_link-shortener will be updated in-place
! resource "keycloak_group_memberships" "project_devops_link-shortener" {
        id       = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935"
!       members  = [
-           "brightz",
            # (10 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_tartan-vote will be updated in-place
! resource "keycloak_group_memberships" "project_tartan-vote" {
        id       = "scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c"
!       members  = [
-           "lwahi",
            # (10 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
-           "lwahi",
-           "lxh",
+           "smwang2",
            # (94 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 8 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 8 to change, 0 to destroy.


3. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.23s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # googleplay_app_iam.eats_stanleymw will be created
+ resource "googleplay_app_iam" "eats_stanleymw" {
+       app_id               = "com.cmueats.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "smwang2@andrew.cmu.edu"
    }

  # googleplay_user.stanleymw will be created
+ resource "googleplay_user" "stanleymw" {
+       email                = "smwang2@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 2 to add, 0 to change, 0 to destroy.


4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.29s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_locho will be destroyed
  # (because google_cloud_identity_group_membership.tech_locho is not in configuration)
- resource "google_cloud_identity_group_membership" "tech_locho" {
-       create_time = "2026-09-27T18:14:27.052016Z" -> null
-       group       = "groups/04anzqyu1bqbv79" -> null
-       id          = "groups/04anzqyu1bqbv79/memberships/111069990805475038275" -> null
-       name        = "groups/04anzqyu1bqbv79/memberships/111069990805475038275" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-27T18:14:27.052016Z" -> null

-       preferred_member_key {
-           id = "lxh@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.tech_lwahi will be destroyed
  # (because google_cloud_identity_group_membership.tech_lwahi is not in configuration)
- resource "google_cloud_identity_group_membership" "tech_lwahi" {
-       create_time = "2026-09-26T23:42:21.456782Z" -> null
-       group       = "groups/04anzqyu1bqbv79" -> null
-       id          = "groups/04anzqyu1bqbv79/memberships/112352542527683783999" -> null
-       name        = "groups/04anzqyu1bqbv79/memberships/112352542527683783999" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-26T23:42:21.456782Z" -> null

-       preferred_member_key {
-           id = "lwahi@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.tech_stanleymw will be created
+ resource "google_cloud_identity_group_membership" "tech_stanleymw" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "smwang2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 2 to destroy.


5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.23s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # forgejo_team_member.cmu-eats_stanleymw will be created
+ resource "forgejo_team_member" "cmu-eats_stanleymw" {
+       team_id = 28
+       user    = "stanleymw"
    }

  # forgejo_team_member.cmu-housing_locho will be destroyed
  # (because forgejo_team_member.cmu-housing_locho is not in configuration)
- resource "forgejo_team_member" "cmu-housing_locho" {
-       team_id = 2 -> null
-       user    = "locho" -> null
    }

  # forgejo_team_member.devops_brightz will be destroyed
  # (because forgejo_team_member.devops_brightz is not in configuration)
- resource "forgejo_team_member" "devops_brightz" {
-       team_id = 4 -> null
-       user    = "brightz" -> null
    }

Plan: 1 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 1 to add, 0 to change, 2 to destroy.


6. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.25s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)
- destroy

OpenTofu will perform the following actions:

  # discord_member_roles.brightz will be updated in-place
! resource "discord_member_roles" "brightz" {
        id        = "1353118209024331948:689114594794668076"
        # (2 unchanged attributes hidden)

-       role {
-           has_role = true -> null
-           role_id  = "1488787087598813215" -> null
        }

        # (2 unchanged blocks hidden)
    }

  # discord_member_roles.locho will be destroyed
  # (because discord_member_roles.locho is not in configuration)
- resource "discord_member_roles" "locho" {
-       id        = "1353118209024331948:582666638550368277" -> null
-       server_id = "1353118209024331948" -> null
-       user_id   = "582666638550368277" -> null

-       role {
-           has_role = true -> null
-           role_id  = "1488787088118644848" -> null
        }
-       role {
-           has_role = true -> null
-           role_id  = "1513653499957874871" -> null
        }
    }

  # discord_member_roles.lwahi will be destroyed
  # (because discord_member_roles.lwahi is not in configuration)
- resource "discord_member_roles" "lwahi" {
-       id        = "1353118209024331948:738509196487688262" -> null
-       server_id = "1353118209024331948" -> null
-       user_id   = "738509196487688262" -> null

-       role {
-           has_role = true -> null
-           role_id  = "1488787088118644848" -> null
        }
-       role {
-           has_role = true -> null
-           role_id  = "1508601984255524885" -> null
        }
    }

  # discord_member_roles.stanleymw will be created
+ resource "discord_member_roles" "stanleymw" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "1367644731571568660"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1541552714063024148"
        }
    }

Plan: 1 to add, 1 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 1 to change, 2 to destroy.


Plan Summary

6 projects, 6 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.17s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # null_resource.slack_cmu-eats_ch0_stanleymw will be created + resource "null_resource" "slack_cmu-eats_ch0_stanleymw" { + id = (known after apply) + triggers = { + "channel" = "C043UV30DHA" + "user" = "U09DUH5550B" } } # null_resource.slack_devops_ch0_brightz will be destroyed # (because null_resource.slack_devops_ch0_brightz is not in configuration) - resource "null_resource" "slack_devops_ch0_brightz" { - id = "1977460316515527708" -> null - triggers = { - "channel" = "C08K3Q77ZQF" - "user" = "U09F3P78LT0" } -> null } # null_resource.slack_devops_ch1_brightz will be destroyed # (because null_resource.slack_devops_ch1_brightz is not in configuration) - resource "null_resource" "slack_devops_ch1_brightz" { - id = "4471079696359905776" -> null - triggers = { - "channel" = "C0BNV84NUTC" - "user" = "U09F3P78LT0" } -> null } # null_resource.slack_hub_locho will be destroyed # (because null_resource.slack_hub_locho is not in configuration) - resource "null_resource" "slack_hub_locho" { - id = "4360225298301891996" -> null - triggers = { - "channel" = "C096TM8EMS8" - "user" = "U0C12EHAGPP" } -> null } # null_resource.slack_hub_lwahi will be destroyed # (because null_resource.slack_hub_lwahi is not in configuration) - resource "null_resource" "slack_hub_lwahi" { - id = "982101409620673204" -> null - triggers = { - "channel" = "C096TM8EMS8" - "user" = "U0C18F0GT7G" } -> null } # null_resource.slack_hub_stanleymw will be created + resource "null_resource" "slack_hub_stanleymw" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U09DUH5550B" } } Plan: 2 to add, 0 to change, 4 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 4 to destroy. --- ### 2. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.26s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-eats_eats will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-eats_eats" { id = "scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba" ! members = [ + "smwang2", # (7 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-housing_housing" { id = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d" ! members = [ - "lxh", # (14 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops will be updated in-place ! resource "keycloak_group_memberships" "project_devops" { id = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7" ! members = [ - "brightz", # (10 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_changelog will be updated in-place ! resource "keycloak_group_memberships" "project_devops_changelog" { id = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77" ! members = [ - "brightz", # (10 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_documentation will be updated in-place ! resource "keycloak_group_memberships" "project_devops_documentation" { id = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f" ! members = [ - "brightz", # (10 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_link-shortener will be updated in-place ! resource "keycloak_group_memberships" "project_devops_link-shortener" { id = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935" ! members = [ - "brightz", # (10 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_tartan-vote will be updated in-place ! resource "keycloak_group_memberships" "project_tartan-vote" { id = "scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c" ! members = [ - "lwahi", # (10 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ - "lwahi", - "lxh", + "smwang2", # (94 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 8 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 8 to change, 0 to destroy. --- ### 3. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.23s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # googleplay_app_iam.eats_stanleymw will be created + resource "googleplay_app_iam" "eats_stanleymw" { + app_id = "com.cmueats.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "smwang2@andrew.cmu.edu" } # googleplay_user.stanleymw will be created + resource "googleplay_user" "stanleymw" { + email = "smwang2@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.29s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_locho will be destroyed # (because google_cloud_identity_group_membership.tech_locho is not in configuration) - resource "google_cloud_identity_group_membership" "tech_locho" { - create_time = "2026-09-27T18:14:27.052016Z" -> null - group = "groups/04anzqyu1bqbv79" -> null - id = "groups/04anzqyu1bqbv79/memberships/111069990805475038275" -> null - name = "groups/04anzqyu1bqbv79/memberships/111069990805475038275" -> null - type = "USER" -> null - update_time = "2026-09-27T18:14:27.052016Z" -> null - preferred_member_key { - id = "lxh@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.tech_lwahi will be destroyed # (because google_cloud_identity_group_membership.tech_lwahi is not in configuration) - resource "google_cloud_identity_group_membership" "tech_lwahi" { - create_time = "2026-09-26T23:42:21.456782Z" -> null - group = "groups/04anzqyu1bqbv79" -> null - id = "groups/04anzqyu1bqbv79/memberships/112352542527683783999" -> null - name = "groups/04anzqyu1bqbv79/memberships/112352542527683783999" -> null - type = "USER" -> null - update_time = "2026-09-26T23:42:21.456782Z" -> null - preferred_member_key { - id = "lwahi@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.tech_stanleymw will be created + resource "google_cloud_identity_group_membership" "tech_stanleymw" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "smwang2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 2 to destroy. --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.23s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # forgejo_team_member.cmu-eats_stanleymw will be created + resource "forgejo_team_member" "cmu-eats_stanleymw" { + team_id = 28 + user = "stanleymw" } # forgejo_team_member.cmu-housing_locho will be destroyed # (because forgejo_team_member.cmu-housing_locho is not in configuration) - resource "forgejo_team_member" "cmu-housing_locho" { - team_id = 2 -> null - user = "locho" -> null } # forgejo_team_member.devops_brightz will be destroyed # (because forgejo_team_member.devops_brightz is not in configuration) - resource "forgejo_team_member" "devops_brightz" { - team_id = 4 -> null - user = "brightz" -> null } Plan: 1 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 1 to add, 0 to change, 2 to destroy. --- ### 6. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.25s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) - destroy OpenTofu will perform the following actions: # discord_member_roles.brightz will be updated in-place ! resource "discord_member_roles" "brightz" { id = "1353118209024331948:689114594794668076" # (2 unchanged attributes hidden) - role { - has_role = true -> null - role_id = "1488787087598813215" -> null } # (2 unchanged blocks hidden) } # discord_member_roles.locho will be destroyed # (because discord_member_roles.locho is not in configuration) - resource "discord_member_roles" "locho" { - id = "1353118209024331948:582666638550368277" -> null - server_id = "1353118209024331948" -> null - user_id = "582666638550368277" -> null - role { - has_role = true -> null - role_id = "1488787088118644848" -> null } - role { - has_role = true -> null - role_id = "1513653499957874871" -> null } } # discord_member_roles.lwahi will be destroyed # (because discord_member_roles.lwahi is not in configuration) - resource "discord_member_roles" "lwahi" { - id = "1353118209024331948:738509196487688262" -> null - server_id = "1353118209024331948" -> null - user_id = "738509196487688262" -> null - role { - has_role = true -> null - role_id = "1488787088118644848" -> null } - role { - has_role = true -> null - role_id = "1508601984255524885" -> null } } # discord_member_roles.stanleymw will be created + resource "discord_member_roles" "stanleymw" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "1367644731571568660" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1541552714063024148" } } Plan: 1 to add, 1 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 1 to change, 2 to destroy. --- ### Plan Summary 6 projects, 6 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
arom1a force-pushed main from abb548b0cf
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/apply Apply failed.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: slack Plan: 2 to add, 0 to change, 4 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 8 to change, 0 to destroy.
atlantis/plan: google_play Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 2 to destroy.
atlantis/plan: git Plan: 1 to add, 0 to change, 2 to destroy.
atlantis/plan: discord Plan: 1 to add, 1 to change, 2 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
to 461eb95499
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_play Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 2 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: discord Apply failed.
atlantis/apply: git Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 1/6 projects failed to apply.
2026-09-27 19:13:23 +00:00
Compare

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: keycloak dir: tofu/keycloak workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: discord dir: tofu/discord workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.27s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-eats_ch0_stanleymw will be created
+ resource "null_resource" "slack_cmu-eats_ch0_stanleymw" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C043UV30DHA"
          + "user"    = "U09DUH5550B"
        }
    }

  # null_resource.slack_hub_stanleymw will be created
+ resource "null_resource" "slack_hub_stanleymw" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U09DUH5550B"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 0 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.10s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # googleplay_app_iam.eats_stanleymw will be created
+ resource "googleplay_app_iam" "eats_stanleymw" {
+       app_id               = "com.cmueats.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "smwang2@andrew.cmu.edu"
    }

  # googleplay_user.stanleymw will be created
+ resource "googleplay_user" "stanleymw" {
+       email                = "smwang2@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 2 to add, 0 to change, 0 to destroy.


3. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.26s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-eats_eats will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-eats_eats" {
        id       = "scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba"
!       members  = [
+           "smwang2",
            # (7 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "smwang2",
            # (96 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 2 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 2 to change, 0 to destroy.


4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.23s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_stanleymw will be created
+ resource "google_cloud_identity_group_membership" "tech_stanleymw" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "smwang2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.19s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # forgejo_team_member.cmu-eats_stanleymw will be created
+ resource "forgejo_team_member" "cmu-eats_stanleymw" {
+       team_id = 28
+       user    = "stanleymw"
    }

  # forgejo_team_member.tartan-vote_lwahi will be created
+ resource "forgejo_team_member" "tartan-vote_lwahi" {
+       team_id = 6
+       user    = "lwahi"
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 0 to change, 0 to destroy.


6. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.stanleymw will be created
+ resource "discord_member_roles" "stanleymw" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "1367644731571568660"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1541552714063024148"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


Plan Summary

6 projects, 6 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.27s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_cmu-eats_ch0_stanleymw will be created + resource "null_resource" "slack_cmu-eats_ch0_stanleymw" { + id = (known after apply) + triggers = { + "channel" = "C043UV30DHA" + "user" = "U09DUH5550B" } } # null_resource.slack_hub_stanleymw will be created + resource "null_resource" "slack_hub_stanleymw" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U09DUH5550B" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.10s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # googleplay_app_iam.eats_stanleymw will be created + resource "googleplay_app_iam" "eats_stanleymw" { + app_id = "com.cmueats.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "smwang2@andrew.cmu.edu" } # googleplay_user.stanleymw will be created + resource "googleplay_user" "stanleymw" { + email = "smwang2@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 3. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.26s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-eats_eats will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-eats_eats" { id = "scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba" ! members = [ + "smwang2", # (7 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "smwang2", # (96 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 2 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 2 to change, 0 to destroy. --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.23s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_stanleymw will be created + resource "google_cloud_identity_group_membership" "tech_stanleymw" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "smwang2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.19s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # forgejo_team_member.cmu-eats_stanleymw will be created + resource "forgejo_team_member" "cmu-eats_stanleymw" { + team_id = 28 + user = "stanleymw" } # forgejo_team_member.tartan-vote_lwahi will be created + resource "forgejo_team_member" "tartan-vote_lwahi" { + team_id = 6 + user = "lwahi" } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 6. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.stanleymw will be created + resource "discord_member_roles" "stanleymw" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "1367644731571568660" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1541552714063024148" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### Plan Summary 6 projects, 6 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
thesuperRL approved these changes 2026-09-27 19:22:27 +00:00
Owner

atlantis apply

atlantis apply

Ran Apply for 6 projects:

  1. project: discord dir: tofu/discord workspace: default
  2. project: git dir: tofu/git workspace: default
  3. project: slack dir: tofu/slack workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: google_play dir: tofu/google_play workspace: default
  6. project: keycloak dir: tofu/keycloak workspace: default

1. project: discord dir: tofu/discord workspace: default

Apply Error

Show Output
running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu apply -input=false /tmp/atlantis-validated-plan-4179203135/plan.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/420/default/tofu/discord': exit status 1
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.stanleymw: Creating...
╷
│ Error: Could not get member 1367644731571568660 in 1353118209024331948: HTTP 404 Not Found, {"message": "Unknown Member", "code": 10007}
│ 
│   with discord_member_roles.stanleymw,
│   on discord.tf.json line 1265, in resource.discord_member_roles.stanleymw:
│ 1265:       },
│ 
╵


2. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.20s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.cmu-eats_stanleymw: Creating...
forgejo_team_member.tartan-vote_lwahi: Creating...
forgejo_team_member.cmu-eats_stanleymw: Creation complete after 0s
forgejo_team_member.tartan-vote_lwahi: Creation complete after 0s

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

3. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.24s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_cmu-eats_ch0_stanleymw: Creating...
null_resource.slack_hub_stanleymw: Creating...
null_resource.slack_cmu-eats_ch0_stanleymw: Provisioning with 'local-exec'...
null_resource.slack_cmu-eats_ch0_stanleymw (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C043UV30DHA --user U09DUH5550B"]
null_resource.slack_hub_stanleymw: Provisioning with 'local-exec'...
null_resource.slack_hub_stanleymw (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U09DUH5550B"]
null_resource.slack_cmu-eats_ch0_stanleymw: Creation complete after 0s [id=5246338771432580315]
null_resource.slack_hub_stanleymw: Creation complete after 0s [id=9161866069573958511]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.28s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_stanleymw: Creating...
google_cloud_identity_group_membership.tech_stanleymw: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/108748397014985502908]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

5. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.17s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
googleplay_user.stanleymw: Creating...
googleplay_user.stanleymw: Creation complete after 1s [id=smwang2@andrew.cmu.edu]
googleplay_app_iam.eats_stanleymw: Creating...
googleplay_app_iam.eats_stanleymw: Creation complete after 1s

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

6. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.34s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cmu-eats_eats: Modifying... [id=scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_cmu-eats_eats: Modifications complete after 2s [id=scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba]
keycloak_group_memberships.projects: Modifications complete after 4s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 2 changed, 0 destroyed.

Apply Summary

6 projects, 5 successful, 0 failed, 1 errored

Ran Apply for 6 projects: 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `discord` dir: `tofu/discord` workspace: `default` **Apply Error** <details><summary>Show Output</summary> ``` running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu apply -input=false /tmp/atlantis-validated-plan-4179203135/plan.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/420/default/tofu/discord': exit status 1 Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.stanleymw: Creating... ╷ │ Error: Could not get member 1367644731571568660 in 1353118209024331948: HTTP 404 Not Found, {"message": "Unknown Member", "code": 10007} │ │ with discord_member_roles.stanleymw, │ on discord.tf.json line 1265, in resource.discord_member_roles.stanleymw: │ 1265: }, │ ╵ ``` </details> --- ### 2. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.20s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.cmu-eats_stanleymw: Creating... forgejo_team_member.tartan-vote_lwahi: Creating... forgejo_team_member.cmu-eats_stanleymw: Creation complete after 0s forgejo_team_member.tartan-vote_lwahi: Creation complete after 0s Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.24s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_cmu-eats_ch0_stanleymw: Creating... null_resource.slack_hub_stanleymw: Creating... null_resource.slack_cmu-eats_ch0_stanleymw: Provisioning with 'local-exec'... null_resource.slack_cmu-eats_ch0_stanleymw (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C043UV30DHA --user U09DUH5550B"] null_resource.slack_hub_stanleymw: Provisioning with 'local-exec'... null_resource.slack_hub_stanleymw (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U09DUH5550B"] null_resource.slack_cmu-eats_ch0_stanleymw: Creation complete after 0s [id=5246338771432580315] null_resource.slack_hub_stanleymw: Creation complete after 0s [id=9161866069573958511] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.28s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_stanleymw: Creating... google_cloud_identity_group_membership.tech_stanleymw: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/108748397014985502908] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.17s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ googleplay_user.stanleymw: Creating... googleplay_user.stanleymw: Creation complete after 1s [id=smwang2@andrew.cmu.edu] googleplay_app_iam.eats_stanleymw: Creating... googleplay_app_iam.eats_stanleymw: Creation complete after 1s Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 6. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.34s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cmu-eats_eats: Modifying... [id=scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_cmu-eats_eats: Modifications complete after 2s [id=scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba] keycloak_group_memberships.projects: Modifications complete after 4s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 2 changed, 0 destroyed. ``` </details> --- ### Apply Summary 6 projects, 5 successful, 0 failed, 1 errored

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_play Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 2 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: discord Apply failed.
atlantis/apply: git Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 1/6 projects failed to apply.
Some workflows are waiting to be reviewed.
Sign in to join this conversation.
No description provided.