chore: adding luluj to slai #402

Closed
luluj wants to merge 2 commits from luluj/governance:main into main
First-time contributor
No description provided.
chore: adding luluj to slai
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 3m25s
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
CI / build (pull_request) Has been cancelled
2a0e8cc1ad
jiyas approved these changes 2026-09-26 20:16:54 +00:00
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 2 to add, 0 to change, 3 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 10 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 1 to destroy.
atlantis/plan: git Plan: 1 to add, 2 to change, 1 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 1 to destroy.
atlantis/plan 5/5 projects planned successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
143e1d9101

Ran Plan for 5 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: keycloak dir: tofu/keycloak workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: git dir: tofu/git workspace: default
  5. project: discord dir: tofu/discord workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.48s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_devops_ch0_junwonp will be destroyed
  # (because null_resource.slack_devops_ch0_junwonp is not in configuration)
- resource "null_resource" "slack_devops_ch0_junwonp" {
-       id       = "8978198533764068893" -> null
-       triggers = {
          - "channel" = "C08K3Q77ZQF"
          - "user"    = "U0C31L4SXAA"
        } -> null
    }

  # null_resource.slack_devops_ch1_junwonp will be destroyed
  # (because null_resource.slack_devops_ch1_junwonp is not in configuration)
- resource "null_resource" "slack_devops_ch1_junwonp" {
-       id       = "9162332391230659035" -> null
-       triggers = {
          - "channel" = "C0BNV84NUTC"
          - "user"    = "U0C31L4SXAA"
        } -> null
    }

  # null_resource.slack_hub_junwonp will be destroyed
  # (because null_resource.slack_hub_junwonp is not in configuration)
- resource "null_resource" "slack_hub_junwonp" {
-       id       = "7012973114601905831" -> null
-       triggers = {
          - "channel" = "C096TM8EMS8"
          - "user"    = "U0C31L4SXAA"
        } -> null
    }

  # null_resource.slack_hub_luluj will be created
+ resource "null_resource" "slack_hub_luluj" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U09FSR7G18A"
        }
    }

  # null_resource.slack_slai_ch0_luluj will be created
+ resource "null_resource" "slack_slai_ch0_luluj" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08SRUGUQ05"
          + "user"    = "U09FSR7G18A"
        }
    }

Plan: 2 to add, 0 to change, 3 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 3 to destroy.


2. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.55s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_devops will be updated in-place
! resource "keycloak_group_memberships" "project_devops" {
        id       = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7"
!       members  = [
-           "junwonp",
            # (9 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_changelog will be updated in-place
! resource "keycloak_group_memberships" "project_devops_changelog" {
        id       = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77"
!       members  = [
-           "junwonp",
            # (9 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_documentation will be updated in-place
! resource "keycloak_group_memberships" "project_devops_documentation" {
        id       = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f"
!       members  = [
-           "junwonp",
            # (9 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_link-shortener will be updated in-place
! resource "keycloak_group_memberships" "project_devops_link-shortener" {
        id       = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935"
!       members  = [
-           "junwonp",
            # (9 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-agent will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-agent" {
        id       = "scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423"
!       members  = [
+           "luluj",
            # (11 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-mcp-server will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server" {
        id       = "scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8"
!       members  = [
+           "luluj",
            # (11 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-sms-surface will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface" {
        id       = "scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43"
!       members  = [
+           "luluj",
            # (11 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-surface will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-surface" {
        id       = "scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775"
!       members  = [
+           "luluj",
            # (11 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_slai-rfcs will be updated in-place
! resource "keycloak_group_memberships" "project_slai_slai-rfcs" {
        id       = "scottylabs/group-memberships/f142f88f-bae2-4480-a343-f170cccbb497"
!       members  = [
+           "luluj",
            # (11 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
-           "junwonp",
+           "luluj",
            # (93 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 10 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 10 to change, 0 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.49s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_junwonp will be destroyed
  # (because google_cloud_identity_group_membership.tech_junwonp is not in configuration)
- resource "google_cloud_identity_group_membership" "tech_junwonp" {
-       create_time = "2026-09-26T20:51:48.130214Z" -> null
-       group       = "groups/04anzqyu1bqbv79" -> null
-       id          = "groups/04anzqyu1bqbv79/memberships/113786424116030298330" -> null
-       name        = "groups/04anzqyu1bqbv79/memberships/113786424116030298330" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-26T20:51:48.130214Z" -> null

-       preferred_member_key {
-           id = "junwonp@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.tech_luluj will be created
+ resource "google_cloud_identity_group_membership" "tech_luluj" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "luluj@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 1 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 1 to destroy.


4. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.50s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)
- destroy

OpenTofu will perform the following actions:

  # forgejo_team_member.devops_junwonp will be destroyed
  # (because forgejo_team_member.devops_junwonp is not in configuration)
- resource "forgejo_team_member" "devops_junwonp" {
-       team_id = 4 -> null
-       user    = "junwonp" -> null
    }

  # forgejo_team_member.slai_luluj will be created
+ resource "forgejo_team_member" "slai_luluj" {
+       team_id = 8
+       user    = "luluj"
    }

  # github_repository.slai_cmugpt_agent will be updated in-place
! resource "github_repository" "slai_cmugpt_agent" {
        id                                      = "cmugpt-agent"
        name                                    = "cmugpt-agent"
        # (39 unchanged attributes hidden)

-       template {
-           include_all_branches = false -> null
-           owner                = "ScottyLabs" -> null
-           repository           = "python-template" -> null
        }

        # (1 unchanged block hidden)
    }

  # github_repository.slai_cmugpt_surface will be updated in-place
! resource "github_repository" "slai_cmugpt_surface" {
        id                                      = "cmugpt-surface"
        name                                    = "cmugpt-surface"
        # (39 unchanged attributes hidden)

-       template {
-           include_all_branches = false -> null
-           owner                = "scottylabs-labrador" -> null
-           repository           = "ScottyStack" -> null
        }

        # (1 unchanged block hidden)
    }

Plan: 1 to add, 2 to change, 1 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 1 to add, 2 to change, 1 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
   Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/402/default/crates/governance-schema)
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/402/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/402/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/402/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 9.46s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # discord_member_roles.junwonp will be destroyed
  # (because discord_member_roles.junwonp is not in configuration)
- resource "discord_member_roles" "junwonp" {
-       id        = "1353118209024331948:662802573467844629" -> null
-       server_id = "1353118209024331948" -> null
-       user_id   = "662802573467844629" -> null

-       role {
-           has_role = true -> null
-           role_id  = "1488787087598813215" -> null
        }
-       role {
-           has_role = true -> null
-           role_id  = "1488787088118644848" -> null
        }
    }

  # discord_member_roles.luluj will be created
+ resource "discord_member_roles" "luluj" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "712406082664005673"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1505078357321191535"
        }
    }

Plan: 1 to add, 0 to change, 1 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 1 to destroy.


Plan Summary

5 projects, 5 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 5 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.48s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # null_resource.slack_devops_ch0_junwonp will be destroyed # (because null_resource.slack_devops_ch0_junwonp is not in configuration) - resource "null_resource" "slack_devops_ch0_junwonp" { - id = "8978198533764068893" -> null - triggers = { - "channel" = "C08K3Q77ZQF" - "user" = "U0C31L4SXAA" } -> null } # null_resource.slack_devops_ch1_junwonp will be destroyed # (because null_resource.slack_devops_ch1_junwonp is not in configuration) - resource "null_resource" "slack_devops_ch1_junwonp" { - id = "9162332391230659035" -> null - triggers = { - "channel" = "C0BNV84NUTC" - "user" = "U0C31L4SXAA" } -> null } # null_resource.slack_hub_junwonp will be destroyed # (because null_resource.slack_hub_junwonp is not in configuration) - resource "null_resource" "slack_hub_junwonp" { - id = "7012973114601905831" -> null - triggers = { - "channel" = "C096TM8EMS8" - "user" = "U0C31L4SXAA" } -> null } # null_resource.slack_hub_luluj will be created + resource "null_resource" "slack_hub_luluj" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U09FSR7G18A" } } # null_resource.slack_slai_ch0_luluj will be created + resource "null_resource" "slack_slai_ch0_luluj" { + id = (known after apply) + triggers = { + "channel" = "C08SRUGUQ05" + "user" = "U09FSR7G18A" } } Plan: 2 to add, 0 to change, 3 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 3 to destroy. --- ### 2. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.55s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_devops will be updated in-place ! resource "keycloak_group_memberships" "project_devops" { id = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7" ! members = [ - "junwonp", # (9 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_changelog will be updated in-place ! resource "keycloak_group_memberships" "project_devops_changelog" { id = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77" ! members = [ - "junwonp", # (9 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_documentation will be updated in-place ! resource "keycloak_group_memberships" "project_devops_documentation" { id = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f" ! members = [ - "junwonp", # (9 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_link-shortener will be updated in-place ! resource "keycloak_group_memberships" "project_devops_link-shortener" { id = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935" ! members = [ - "junwonp", # (9 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-agent will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-agent" { id = "scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423" ! members = [ + "luluj", # (11 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-mcp-server will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server" { id = "scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8" ! members = [ + "luluj", # (11 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-sms-surface will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface" { id = "scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43" ! members = [ + "luluj", # (11 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-surface will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-surface" { id = "scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775" ! members = [ + "luluj", # (11 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_slai-rfcs will be updated in-place ! resource "keycloak_group_memberships" "project_slai_slai-rfcs" { id = "scottylabs/group-memberships/f142f88f-bae2-4480-a343-f170cccbb497" ! members = [ + "luluj", # (11 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ - "junwonp", + "luluj", # (93 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 10 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 10 to change, 0 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.49s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_junwonp will be destroyed # (because google_cloud_identity_group_membership.tech_junwonp is not in configuration) - resource "google_cloud_identity_group_membership" "tech_junwonp" { - create_time = "2026-09-26T20:51:48.130214Z" -> null - group = "groups/04anzqyu1bqbv79" -> null - id = "groups/04anzqyu1bqbv79/memberships/113786424116030298330" -> null - name = "groups/04anzqyu1bqbv79/memberships/113786424116030298330" -> null - type = "USER" -> null - update_time = "2026-09-26T20:51:48.130214Z" -> null - preferred_member_key { - id = "junwonp@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.tech_luluj will be created + resource "google_cloud_identity_group_membership" "tech_luluj" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "luluj@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 1 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 1 to destroy. --- ### 4. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.50s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) - destroy OpenTofu will perform the following actions: # forgejo_team_member.devops_junwonp will be destroyed # (because forgejo_team_member.devops_junwonp is not in configuration) - resource "forgejo_team_member" "devops_junwonp" { - team_id = 4 -> null - user = "junwonp" -> null } # forgejo_team_member.slai_luluj will be created + resource "forgejo_team_member" "slai_luluj" { + team_id = 8 + user = "luluj" } # github_repository.slai_cmugpt_agent will be updated in-place ! resource "github_repository" "slai_cmugpt_agent" { id = "cmugpt-agent" name = "cmugpt-agent" # (39 unchanged attributes hidden) - template { - include_all_branches = false -> null - owner = "ScottyLabs" -> null - repository = "python-template" -> null } # (1 unchanged block hidden) } # github_repository.slai_cmugpt_surface will be updated in-place ! resource "github_repository" "slai_cmugpt_surface" { id = "cmugpt-surface" name = "cmugpt-surface" # (39 unchanged attributes hidden) - template { - include_all_branches = false -> null - owner = "scottylabs-labrador" -> null - repository = "ScottyStack" -> null } # (1 unchanged block hidden) } Plan: 1 to add, 2 to change, 1 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 1 to add, 2 to change, 1 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/402/default/crates/governance-schema) Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/402/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/402/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/402/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 9.46s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # discord_member_roles.junwonp will be destroyed # (because discord_member_roles.junwonp is not in configuration) - resource "discord_member_roles" "junwonp" { - id = "1353118209024331948:662802573467844629" -> null - server_id = "1353118209024331948" -> null - user_id = "662802573467844629" -> null - role { - has_role = true -> null - role_id = "1488787087598813215" -> null } - role { - has_role = true -> null - role_id = "1488787088118644848" -> null } } # discord_member_roles.luluj will be created + resource "discord_member_roles" "luluj" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "712406082664005673" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1505078357321191535" } } Plan: 1 to add, 0 to change, 1 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 1 to destroy. --- ### Plan Summary 5 projects, 5 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

superseded by #422

superseded by https://git.cmu.dev/ScottyLabs/governance/pulls/422
anish closed this pull request 2026-09-26 22:11:26 +00:00
Member

atlantis unlock

atlantis unlock

All Atlantis locks for this PR have been unlocked and plans discarded

All Atlantis locks for this PR have been unlocked and plans discarded
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 2 to add, 0 to change, 3 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 10 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 1 to destroy.
atlantis/plan: git Plan: 1 to add, 2 to change, 1 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 1 to destroy.
atlantis/plan 5/5 projects planned successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.

Pull request closed

Sign in to join this conversation.
No description provided.