feat: add "sypan" to maps #412
No reviewers
Labels
No labels
needs/linking
needs/docs
needs/research
needs/upstream
type/bug
type/external
type/feature
type/refactor
urgency
high
urgency
immediate
urgency
low
urgency
medium
No milestone
No project
No assignees
4 participants
Notifications
Due date
No due date set.
Dependencies
No dependencies set
Reference
ScottyLabs/governance!412
Loading…
Reference in a new issue
No description provided.
Delete branch "sypan/governance:join-cmu_maps"
Deleting a branch is permanent. Although the deleted branch may continue to exist for a short time before it actually gets removed, it CANNOT be undone in most cases. Continue?
Ran Plan for 6 projects:
gitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgoogle_playdir:tofu/google_playworkspace:defaultslackdir:tofu/slackworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:default1. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #402. To continue, delete the lock from #402 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #402. To continue, delete the lock from #402 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
google_playdir:tofu/google_playworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #402. To continue, delete the lock from #402 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #402. To continue, delete the lock from #402 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.6. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #402. To continue, delete the lock from #402 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.Plan Summary
6 projects, 0 with changes, 0 with no changes, 6 failed
atlantis plan
atlantis plan
atlantis plan
atlantis plan
Ran Plan for 6 projects:
slackdir:tofu/slackworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultgoogle_playdir:tofu/google_playworkspace:defaultgitdir:tofu/gitworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultdiscorddir:tofu/discordworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
google_playdir:tofu/google_playworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #262. To continue, delete the lock from #262 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.6. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.Plan Summary
6 projects, 0 with changes, 0 with no changes, 6 failed
View command line instructions
Manual merge helper
Use this merge commit message when completing the merge manually.
Checkout
From your project repository, check out a new branch and test the changes.Merge
Merge the changes and update on Forgejo.Warning: The "Autodetect manual merge" setting is not enabled for this repository, you will have to mark this pull request as manually merged afterwards.