feat: added jhhan2 to devops #328

Merged
anish merged 1 commit from jhhan2/governance:main into main 2026-09-22 12:54:45 +00:00
Member
No description provided.
feat: added jhhan2 to devops
Some checks failed
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
CI / build (pull_request) Has been cancelled
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 2m50s
ee32fb3de0
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: git Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan 6/6 projects failed to plan.
kennel/build skipped: no services or static sites defined
7c4c33deb5

Ran Plan for 6 projects:

  1. project: discord dir: tofu/discord workspace: default
  2. project: git dir: tofu/git workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: google_play dir: tofu/google_play workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default
  6. project: slack dir: tofu/slack workspace: default

1. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` --- ### 1. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
anish force-pushed main from 7c4c33deb5
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: git Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan 6/6 projects failed to plan.
kennel/build skipped: no services or static sites defined
to 670ac05551
Some checks failed
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
Validate / Governance (pull_request_target) Has been cancelled
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
atlantis/plan: slack Plan: 3 to add, 0 to change, 1 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 6 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 1 to destroy.
atlantis/plan: git Plan: 1 to add, 0 to change, 2 to destroy.
atlantis/plan: discord Plan: 1 to add, 1 to change, 0 to destroy.
atlantis/plan 5/5 projects planned successfully.
2026-09-22 12:38:25 +00:00
Compare
anish force-pushed main from 670ac05551
Some checks failed
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
Validate / Governance (pull_request_target) Has been cancelled
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
atlantis/plan: slack Plan: 3 to add, 0 to change, 1 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 6 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 1 to destroy.
atlantis/plan: git Plan: 1 to add, 0 to change, 2 to destroy.
atlantis/plan: discord Plan: 1 to add, 1 to change, 0 to destroy.
atlantis/plan 5/5 projects planned successfully.
to 56e6b8d872
Some checks failed
CI / build (pull_request) Has been skipped
atlantis/plan: slack Plan: 3 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 5 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 1 to add, 0 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan 5/5 projects planned successfully.
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
Validate / Governance (pull_request_target) Has been cancelled
2026-09-22 12:38:39 +00:00
Compare

Ran Plan for 5 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: keycloak dir: tofu/keycloak workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: git dir: tofu/git workspace: default
  5. project: discord dir: tofu/discord workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/328/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/328/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/328/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 7.65s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_devops_ch0_jhhan2 will be created
+ resource "null_resource" "slack_devops_ch0_jhhan2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08K3Q77ZQF"
          + "user"    = "U0C1FDWSRH8"
        }
    }

  # null_resource.slack_devops_ch1_jhhan2 will be created
+ resource "null_resource" "slack_devops_ch1_jhhan2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0BNV84NUTC"
          + "user"    = "U0C1FDWSRH8"
        }
    }

  # null_resource.slack_hub_jhhan2 will be created
+ resource "null_resource" "slack_hub_jhhan2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0C1FDWSRH8"
        }
    }

  # null_resource.slack_leads_cseluzhy will be destroyed
  # (because null_resource.slack_leads_cseluzhy is not in configuration)
- resource "null_resource" "slack_leads_cseluzhy" {
-       id       = "662549664703465762" -> null
-       triggers = {
          - "channel" = "C0AN1M0HDV0"
          - "user"    = "U09G9PKKLG2"
        } -> null
    }

Plan: 3 to add, 0 to change, 1 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 3 to add, 0 to change, 1 to destroy.


2. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.68s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-study_study_admins will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-study_study_admins" {
        id       = "scottylabs/group-memberships/a683866f-c161-4298-8d00-911d611e2895"
!       members  = [
-           "cseluzhy",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops will be updated in-place
! resource "keycloak_group_memberships" "project_devops" {
        id       = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7"
!       members  = [
+           "jhhan2",
            # (8 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_changelog will be updated in-place
! resource "keycloak_group_memberships" "project_devops_changelog" {
        id       = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77"
!       members  = [
+           "jhhan2",
            # (8 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_documentation will be updated in-place
! resource "keycloak_group_memberships" "project_devops_documentation" {
        id       = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f"
!       members  = [
+           "jhhan2",
            # (8 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_link-shortener will be updated in-place
! resource "keycloak_group_memberships" "project_devops_link-shortener" {
        id       = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935"
!       members  = [
+           "jhhan2",
            # (8 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "jhhan2",
            # (77 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 6 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 6 to change, 0 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.65s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.ops_cseluzhy will be destroyed
  # (because google_cloud_identity_group_membership.ops_cseluzhy is not in configuration)
- resource "google_cloud_identity_group_membership" "ops_cseluzhy" {
-       create_time = "2026-09-22T12:23:01.953138Z" -> null
-       group       = "groups/03fwokq036zsp1f" -> null
-       id          = "groups/03fwokq036zsp1f/memberships/109508502479832362677" -> null
-       name        = "groups/03fwokq036zsp1f/memberships/109508502479832362677" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-22T12:23:01.953138Z" -> null

-       preferred_member_key {
-           id = "cseluzhy@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.tech_jhhan2 will be created
+ resource "google_cloud_identity_group_membership" "tech_jhhan2" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "jhhan2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 1 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 1 to destroy.


4. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.65s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # forgejo_team_member.devops_jhhan2 will be created
+ resource "forgejo_team_member" "devops_jhhan2" {
+       team_id = 4
+       user    = "jhhan2"
    }

  # forgejo_team_member.study_leads_cseluzhy will be destroyed
  # (because forgejo_team_member.study_leads_cseluzhy is not in configuration)
- resource "forgejo_team_member" "study_leads_cseluzhy" {
-       team_id = 42 -> null
-       user    = "cseluzhy" -> null
    }

  # forgejo_team_member.tech_leads_cseluzhy will be destroyed
  # (because forgejo_team_member.tech_leads_cseluzhy is not in configuration)
- resource "forgejo_team_member" "tech_leads_cseluzhy" {
-       team_id = 44 -> null
-       user    = "cseluzhy" -> null
    }

Plan: 1 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 1 to add, 0 to change, 2 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.65s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # discord_member_roles.cseluzhy will be updated in-place
! resource "discord_member_roles" "cseluzhy" {
        id        = "1353118209024331948:1128028390365085747"
        # (2 unchanged attributes hidden)

-       role {
-           has_role = true -> null
-           role_id  = "1519792320755662928" -> null
        }

        # (2 unchanged blocks hidden)
    }

  # discord_member_roles.jhhan2 will be created
+ resource "discord_member_roles" "jhhan2" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "630616046285619210"

+       role {
+           has_role = true
+           role_id  = "1488787087598813215"
        }
+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
    }

Plan: 1 to add, 1 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 1 to change, 0 to destroy.


Plan Summary

5 projects, 5 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 5 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/328/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/328/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/328/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 7.65s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # null_resource.slack_devops_ch0_jhhan2 will be created + resource "null_resource" "slack_devops_ch0_jhhan2" { + id = (known after apply) + triggers = { + "channel" = "C08K3Q77ZQF" + "user" = "U0C1FDWSRH8" } } # null_resource.slack_devops_ch1_jhhan2 will be created + resource "null_resource" "slack_devops_ch1_jhhan2" { + id = (known after apply) + triggers = { + "channel" = "C0BNV84NUTC" + "user" = "U0C1FDWSRH8" } } # null_resource.slack_hub_jhhan2 will be created + resource "null_resource" "slack_hub_jhhan2" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0C1FDWSRH8" } } # null_resource.slack_leads_cseluzhy will be destroyed # (because null_resource.slack_leads_cseluzhy is not in configuration) - resource "null_resource" "slack_leads_cseluzhy" { - id = "662549664703465762" -> null - triggers = { - "channel" = "C0AN1M0HDV0" - "user" = "U09G9PKKLG2" } -> null } Plan: 3 to add, 0 to change, 1 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 3 to add, 0 to change, 1 to destroy. --- ### 2. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.68s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-study_study_admins will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-study_study_admins" { id = "scottylabs/group-memberships/a683866f-c161-4298-8d00-911d611e2895" ! members = [ - "cseluzhy", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops will be updated in-place ! resource "keycloak_group_memberships" "project_devops" { id = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7" ! members = [ + "jhhan2", # (8 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_changelog will be updated in-place ! resource "keycloak_group_memberships" "project_devops_changelog" { id = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77" ! members = [ + "jhhan2", # (8 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_documentation will be updated in-place ! resource "keycloak_group_memberships" "project_devops_documentation" { id = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f" ! members = [ + "jhhan2", # (8 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_link-shortener will be updated in-place ! resource "keycloak_group_memberships" "project_devops_link-shortener" { id = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935" ! members = [ + "jhhan2", # (8 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "jhhan2", # (77 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 6 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 6 to change, 0 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.65s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # google_cloud_identity_group_membership.ops_cseluzhy will be destroyed # (because google_cloud_identity_group_membership.ops_cseluzhy is not in configuration) - resource "google_cloud_identity_group_membership" "ops_cseluzhy" { - create_time = "2026-09-22T12:23:01.953138Z" -> null - group = "groups/03fwokq036zsp1f" -> null - id = "groups/03fwokq036zsp1f/memberships/109508502479832362677" -> null - name = "groups/03fwokq036zsp1f/memberships/109508502479832362677" -> null - type = "USER" -> null - update_time = "2026-09-22T12:23:01.953138Z" -> null - preferred_member_key { - id = "cseluzhy@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.tech_jhhan2 will be created + resource "google_cloud_identity_group_membership" "tech_jhhan2" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "jhhan2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 1 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 1 to destroy. --- ### 4. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.65s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # forgejo_team_member.devops_jhhan2 will be created + resource "forgejo_team_member" "devops_jhhan2" { + team_id = 4 + user = "jhhan2" } # forgejo_team_member.study_leads_cseluzhy will be destroyed # (because forgejo_team_member.study_leads_cseluzhy is not in configuration) - resource "forgejo_team_member" "study_leads_cseluzhy" { - team_id = 42 -> null - user = "cseluzhy" -> null } # forgejo_team_member.tech_leads_cseluzhy will be destroyed # (because forgejo_team_member.tech_leads_cseluzhy is not in configuration) - resource "forgejo_team_member" "tech_leads_cseluzhy" { - team_id = 44 -> null - user = "cseluzhy" -> null } Plan: 1 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 1 to add, 0 to change, 2 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.65s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # discord_member_roles.cseluzhy will be updated in-place ! resource "discord_member_roles" "cseluzhy" { id = "1353118209024331948:1128028390365085747" # (2 unchanged attributes hidden) - role { - has_role = true -> null - role_id = "1519792320755662928" -> null } # (2 unchanged blocks hidden) } # discord_member_roles.jhhan2 will be created + resource "discord_member_roles" "jhhan2" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "630616046285619210" + role { + has_role = true + role_id = "1488787087598813215" } + role { + has_role = true + role_id = "1488787088118644848" } } Plan: 1 to add, 1 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 1 to change, 0 to destroy. --- ### Plan Summary 5 projects, 5 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```

Ran Plan for 5 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: keycloak dir: tofu/keycloak workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: git dir: tofu/git workspace: default
  5. project: discord dir: tofu/discord workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.17s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_devops_ch0_jhhan2 will be created
+ resource "null_resource" "slack_devops_ch0_jhhan2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08K3Q77ZQF"
          + "user"    = "U0C1FDWSRH8"
        }
    }

  # null_resource.slack_devops_ch1_jhhan2 will be created
+ resource "null_resource" "slack_devops_ch1_jhhan2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0BNV84NUTC"
          + "user"    = "U0C1FDWSRH8"
        }
    }

  # null_resource.slack_hub_jhhan2 will be created
+ resource "null_resource" "slack_hub_jhhan2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0C1FDWSRH8"
        }
    }

Plan: 3 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 3 to add, 0 to change, 0 to destroy.


2. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.19s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_devops will be updated in-place
! resource "keycloak_group_memberships" "project_devops" {
        id       = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7"
!       members  = [
+           "jhhan2",
            # (8 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_changelog will be updated in-place
! resource "keycloak_group_memberships" "project_devops_changelog" {
        id       = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77"
!       members  = [
+           "jhhan2",
            # (8 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_documentation will be updated in-place
! resource "keycloak_group_memberships" "project_devops_documentation" {
        id       = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f"
!       members  = [
+           "jhhan2",
            # (8 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_link-shortener will be updated in-place
! resource "keycloak_group_memberships" "project_devops_link-shortener" {
        id       = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935"
!       members  = [
+           "jhhan2",
            # (8 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "jhhan2",
            # (77 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 5 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 5 to change, 0 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.17s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_jhhan2 will be created
+ resource "google_cloud_identity_group_membership" "tech_jhhan2" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "jhhan2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


4. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.16s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # forgejo_team_member.devops_jhhan2 will be created
+ resource "forgejo_team_member" "devops_jhhan2" {
+       team_id = 4
+       user    = "jhhan2"
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 1 to add, 0 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.18s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.jhhan2 will be created
+ resource "discord_member_roles" "jhhan2" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "630616046285619210"

+       role {
+           has_role = true
+           role_id  = "1488787087598813215"
        }
+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


Plan Summary

5 projects, 5 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 5 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.17s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_devops_ch0_jhhan2 will be created + resource "null_resource" "slack_devops_ch0_jhhan2" { + id = (known after apply) + triggers = { + "channel" = "C08K3Q77ZQF" + "user" = "U0C1FDWSRH8" } } # null_resource.slack_devops_ch1_jhhan2 will be created + resource "null_resource" "slack_devops_ch1_jhhan2" { + id = (known after apply) + triggers = { + "channel" = "C0BNV84NUTC" + "user" = "U0C1FDWSRH8" } } # null_resource.slack_hub_jhhan2 will be created + resource "null_resource" "slack_hub_jhhan2" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0C1FDWSRH8" } } Plan: 3 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 3 to add, 0 to change, 0 to destroy. --- ### 2. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.19s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_devops will be updated in-place ! resource "keycloak_group_memberships" "project_devops" { id = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7" ! members = [ + "jhhan2", # (8 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_changelog will be updated in-place ! resource "keycloak_group_memberships" "project_devops_changelog" { id = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77" ! members = [ + "jhhan2", # (8 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_documentation will be updated in-place ! resource "keycloak_group_memberships" "project_devops_documentation" { id = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f" ! members = [ + "jhhan2", # (8 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_link-shortener will be updated in-place ! resource "keycloak_group_memberships" "project_devops_link-shortener" { id = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935" ! members = [ + "jhhan2", # (8 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "jhhan2", # (77 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 5 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 5 to change, 0 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.17s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_jhhan2 will be created + resource "google_cloud_identity_group_membership" "tech_jhhan2" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "jhhan2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 4. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.16s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # forgejo_team_member.devops_jhhan2 will be created + resource "forgejo_team_member" "devops_jhhan2" { + team_id = 4 + user = "jhhan2" } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.18s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.jhhan2 will be created + resource "discord_member_roles" "jhhan2" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "630616046285619210" + role { + has_role = true + role_id = "1488787087598813215" } + role { + has_role = true + role_id = "1488787088118644848" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### Plan Summary 5 projects, 5 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis unlock

atlantis unlock

All Atlantis locks for this PR have been unlocked and plans discarded

All Atlantis locks for this PR have been unlocked and plans discarded
anish force-pushed main from 56e6b8d872
Some checks failed
CI / build (pull_request) Has been skipped
atlantis/plan: slack Plan: 3 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 5 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 1 to add, 0 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan 5/5 projects planned successfully.
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
Validate / Governance (pull_request_target) Has been cancelled
to 8bf2c0fa7a
All checks were successful
CI / build (pull_request) Has been skipped
atlantis/plan: slack Plan: 3 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 5 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan 5/5 projects planned successfully.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 4m28s
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 5/5 projects applied successfully.
CI / check-1 (pull_request) Successful in 4m58s
CI / check (pull_request) Successful in 0s
2026-09-22 12:41:09 +00:00
Compare

Ran Plan for 5 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: keycloak dir: tofu/keycloak workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: git dir: tofu/git workspace: default
  5. project: discord dir: tofu/discord workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.13s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_devops_ch0_jhhan2 will be created
+ resource "null_resource" "slack_devops_ch0_jhhan2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08K3Q77ZQF"
          + "user"    = "U0C1FDWSRH8"
        }
    }

  # null_resource.slack_devops_ch1_jhhan2 will be created
+ resource "null_resource" "slack_devops_ch1_jhhan2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0BNV84NUTC"
          + "user"    = "U0C1FDWSRH8"
        }
    }

  # null_resource.slack_hub_jhhan2 will be created
+ resource "null_resource" "slack_hub_jhhan2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0C1FDWSRH8"
        }
    }

Plan: 3 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 3 to add, 0 to change, 0 to destroy.


2. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.16s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_devops will be updated in-place
! resource "keycloak_group_memberships" "project_devops" {
        id       = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7"
!       members  = [
+           "jhhan2",
            # (8 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_changelog will be updated in-place
! resource "keycloak_group_memberships" "project_devops_changelog" {
        id       = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77"
!       members  = [
+           "jhhan2",
            # (8 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_documentation will be updated in-place
! resource "keycloak_group_memberships" "project_devops_documentation" {
        id       = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f"
!       members  = [
+           "jhhan2",
            # (8 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_link-shortener will be updated in-place
! resource "keycloak_group_memberships" "project_devops_link-shortener" {
        id       = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935"
!       members  = [
+           "jhhan2",
            # (8 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "jhhan2",
            # (79 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 5 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 5 to change, 0 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.23s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_jhhan2 will be created
+ resource "google_cloud_identity_group_membership" "tech_jhhan2" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "jhhan2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


4. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.19s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # forgejo_team_member.devops_jhhan2 will be created
+ resource "forgejo_team_member" "devops_jhhan2" {
+       team_id = 4
+       user    = "jhhan2"
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 1 to add, 0 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.26s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.jhhan2 will be created
+ resource "discord_member_roles" "jhhan2" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "630616046285619210"

+       role {
+           has_role = true
+           role_id  = "1488787087598813215"
        }
+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


Plan Summary

5 projects, 5 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 5 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.13s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_devops_ch0_jhhan2 will be created + resource "null_resource" "slack_devops_ch0_jhhan2" { + id = (known after apply) + triggers = { + "channel" = "C08K3Q77ZQF" + "user" = "U0C1FDWSRH8" } } # null_resource.slack_devops_ch1_jhhan2 will be created + resource "null_resource" "slack_devops_ch1_jhhan2" { + id = (known after apply) + triggers = { + "channel" = "C0BNV84NUTC" + "user" = "U0C1FDWSRH8" } } # null_resource.slack_hub_jhhan2 will be created + resource "null_resource" "slack_hub_jhhan2" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0C1FDWSRH8" } } Plan: 3 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 3 to add, 0 to change, 0 to destroy. --- ### 2. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.16s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_devops will be updated in-place ! resource "keycloak_group_memberships" "project_devops" { id = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7" ! members = [ + "jhhan2", # (8 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_changelog will be updated in-place ! resource "keycloak_group_memberships" "project_devops_changelog" { id = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77" ! members = [ + "jhhan2", # (8 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_documentation will be updated in-place ! resource "keycloak_group_memberships" "project_devops_documentation" { id = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f" ! members = [ + "jhhan2", # (8 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_link-shortener will be updated in-place ! resource "keycloak_group_memberships" "project_devops_link-shortener" { id = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935" ! members = [ + "jhhan2", # (8 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "jhhan2", # (79 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 5 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 5 to change, 0 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.23s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_jhhan2 will be created + resource "google_cloud_identity_group_membership" "tech_jhhan2" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "jhhan2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 4. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.19s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # forgejo_team_member.devops_jhhan2 will be created + resource "forgejo_team_member" "devops_jhhan2" { + team_id = 4 + user = "jhhan2" } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.26s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.jhhan2 will be created + resource "discord_member_roles" "jhhan2" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "630616046285619210" + role { + has_role = true + role_id = "1488787087598813215" } + role { + has_role = true + role_id = "1488787088118644848" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### Plan Summary 5 projects, 5 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis apply

atlantis apply

Ran Apply for 5 projects:

  1. project: git dir: tofu/git workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default

1. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.devops_jhhan2: Creating...
forgejo_team_member.devops_jhhan2: Creation complete after 0s

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

2. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_devops_ch0_jhhan2: Creating...
null_resource.slack_devops_ch1_jhhan2: Creating...
null_resource.slack_hub_jhhan2: Creating...
null_resource.slack_devops_ch0_jhhan2: Provisioning with 'local-exec'...
null_resource.slack_devops_ch1_jhhan2: Provisioning with 'local-exec'...
null_resource.slack_devops_ch0_jhhan2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08K3Q77ZQF --user U0C1FDWSRH8"]
null_resource.slack_devops_ch1_jhhan2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0BNV84NUTC --user U0C1FDWSRH8"]
null_resource.slack_hub_jhhan2: Provisioning with 'local-exec'...
null_resource.slack_hub_jhhan2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C1FDWSRH8"]
null_resource.slack_hub_jhhan2: Creation complete after 0s [id=333305548515570789]
null_resource.slack_devops_ch0_jhhan2: Creation complete after 0s [id=5440140062483976153]
null_resource.slack_devops_ch1_jhhan2: Creation complete after 0s [id=9112871001818787871]

Apply complete! Resources: 3 added, 0 changed, 0 destroyed.

3. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.17s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.jhhan2: Creating...
discord_member_roles.jhhan2: Creation complete after 1s [id=1353118209024331948:630616046285619210]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.14s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_jhhan2: Creating...
google_cloud_identity_group_membership.tech_jhhan2: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/106960203739933866519]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

5. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.13s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_devops_documentation: Modifying... [id=scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f]
keycloak_group_memberships.project_devops_link-shortener: Modifying... [id=scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935]
keycloak_group_memberships.project_devops_changelog: Modifying... [id=scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77]
keycloak_group_memberships.project_devops: Modifying... [id=scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_devops_documentation: Modifications complete after 2s [id=scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f]
keycloak_group_memberships.project_devops_changelog: Modifications complete after 2s [id=scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77]
keycloak_group_memberships.project_devops: Modifications complete after 2s [id=scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7]
keycloak_group_memberships.project_devops_link-shortener: Modifications complete after 2s [id=scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935]
keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 5 changed, 0 destroyed.

Apply Summary

5 projects, 5 successful, 0 failed, 0 errored

Ran Apply for 5 projects: 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.devops_jhhan2: Creating... forgejo_team_member.devops_jhhan2: Creation complete after 0s Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_devops_ch0_jhhan2: Creating... null_resource.slack_devops_ch1_jhhan2: Creating... null_resource.slack_hub_jhhan2: Creating... null_resource.slack_devops_ch0_jhhan2: Provisioning with 'local-exec'... null_resource.slack_devops_ch1_jhhan2: Provisioning with 'local-exec'... null_resource.slack_devops_ch0_jhhan2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08K3Q77ZQF --user U0C1FDWSRH8"] null_resource.slack_devops_ch1_jhhan2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0BNV84NUTC --user U0C1FDWSRH8"] null_resource.slack_hub_jhhan2: Provisioning with 'local-exec'... null_resource.slack_hub_jhhan2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C1FDWSRH8"] null_resource.slack_hub_jhhan2: Creation complete after 0s [id=333305548515570789] null_resource.slack_devops_ch0_jhhan2: Creation complete after 0s [id=5440140062483976153] null_resource.slack_devops_ch1_jhhan2: Creation complete after 0s [id=9112871001818787871] Apply complete! Resources: 3 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.17s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.jhhan2: Creating... discord_member_roles.jhhan2: Creation complete after 1s [id=1353118209024331948:630616046285619210] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.14s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_jhhan2: Creating... google_cloud_identity_group_membership.tech_jhhan2: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/106960203739933866519] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.13s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_devops_documentation: Modifying... [id=scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f] keycloak_group_memberships.project_devops_link-shortener: Modifying... [id=scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935] keycloak_group_memberships.project_devops_changelog: Modifying... [id=scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77] keycloak_group_memberships.project_devops: Modifying... [id=scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_devops_documentation: Modifications complete after 2s [id=scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f] keycloak_group_memberships.project_devops_changelog: Modifications complete after 2s [id=scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77] keycloak_group_memberships.project_devops: Modifications complete after 2s [id=scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7] keycloak_group_memberships.project_devops_link-shortener: Modifications complete after 2s [id=scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935] keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 5 changed, 0 destroyed. ``` </details> --- ### Apply Summary 5 projects, 5 successful, 0 failed, 0 errored
anish merged commit af2005a59f into main 2026-09-22 12:54:45 +00:00
anish referenced this pull request from a commit 2026-09-22 12:54:47 +00:00

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
brightz referenced this pull request from a commit 2026-09-26 22:03:28 +00:00
Sign in to join this conversation.
No description provided.