feat: add ezhu1057, jcz2, jliao2 to slai #342

Merged
anish merged 2 commits from add-slai-unapproved into main 2026-09-22 12:41:00 +00:00
Owner

Consolidates unapproved: #202 #274 #325

Consolidates unapproved: #202 #274 #325
feat: add ezhu1057, jcz2, jliao2 to slai (unapproved)
Some checks failed
atlantis/plan Plan failed.
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers failed.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Failing after 12m43s
CI / check-1 (pull_request) Failing after 15m14s
CI / check (pull_request) Failing after 0s
3903c41629
Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/342/default": 
Download attempt 1 failed, retrying...
Download attempt 2 failed, retrying...
Download attempt 3 failed, retrying...
Failed to download forgejo provider after 3 attempts

``` Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/342/default": Download attempt 1 failed, retrying... Download attempt 2 failed, retrying... Download attempt 3 failed, retrying... Failed to download forgejo provider after 3 attempts ```
thesuperRL force-pushed add-slai-unapproved from 3903c41629
Some checks failed
atlantis/plan Plan failed.
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers failed.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Failing after 12m43s
CI / check-1 (pull_request) Failing after 15m14s
CI / check (pull_request) Failing after 0s
to 4d88a2579b
Some checks failed
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Failing after 10m20s
CI / check-1 (pull_request) Successful in 11m40s
CI / check (pull_request) Successful in 0s
2026-09-21 21:59:43 +00:00
Compare
anish force-pushed add-slai-unapproved from 4d88a2579b
Some checks failed
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Failing after 10m20s
CI / check-1 (pull_request) Successful in 11m40s
CI / check (pull_request) Successful in 0s
to 599dbd8aa1
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
CI / build (pull_request) Has been skipped
Validate / Governance (pull_request_target) Successful in 2m51s
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
2026-09-22 12:23:57 +00:00
Compare
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 4 to add, 0 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 5 to change, 0 to destroy.
atlantis/plan: git Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: discord Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan 5/5 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 5/5 projects applied successfully.
d5c44d97df

Ran Plan for 5 projects:

  1. project: discord dir: tofu/discord workspace: default
  2. project: keycloak dir: tofu/keycloak workspace: default
  3. project: slack dir: tofu/slack workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: git dir: tofu/git workspace: default

1. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #328. To continue, delete the lock from #328 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #328. To continue, delete the lock from #328 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #328. To continue, delete the lock from #328 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #328. To continue, delete the lock from #328 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #328. To continue, delete the lock from #328 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

5 projects, 0 with changes, 0 with no changes, 5 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 5 projects: 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #328. To continue, delete the lock from #328 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #328. To continue, delete the lock from #328 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #328. To continue, delete the lock from #328 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #328. To continue, delete the lock from #328 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #328. To continue, delete the lock from #328 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 5 projects, 0 with changes, 0 with no changes, 5 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis plan

atlantis plan

Ran Plan for 5 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_groups dir: tofu/google_groups workspace: default
  3. project: keycloak dir: tofu/keycloak workspace: default
  4. project: git dir: tofu/git workspace: default
  5. project: discord dir: tofu/discord workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.20s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_hub_jcz2 will be created
+ resource "null_resource" "slack_hub_jcz2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0BTHUYABJP"
        }
    }

  # null_resource.slack_hub_jliao2 will be created
+ resource "null_resource" "slack_hub_jliao2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U09F23EKK6X"
        }
    }

  # null_resource.slack_slai_ch0_jcz2 will be created
+ resource "null_resource" "slack_slai_ch0_jcz2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08SRUGUQ05"
          + "user"    = "U0BTHUYABJP"
        }
    }

  # null_resource.slack_slai_ch0_jliao2 will be created
+ resource "null_resource" "slack_slai_ch0_jliao2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08SRUGUQ05"
          + "user"    = "U09F23EKK6X"
        }
    }

Plan: 4 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 4 to add, 0 to change, 0 to destroy.


2. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.20s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_jcz2 will be created
+ resource "google_cloud_identity_group_membership" "tech_jcz2" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "jcz2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

  # google_cloud_identity_group_membership.tech_jliao2 will be created
+ resource "google_cloud_identity_group_membership" "tech_jliao2" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "jliao2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 2 to add, 0 to change, 0 to destroy.


3. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.16s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_slai_cmugpt-agent will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-agent" {
        id       = "scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423"
!       members  = [
+           "jcz2",
+           "jliao2",
            # (9 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-mcp-server will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server" {
        id       = "scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8"
!       members  = [
+           "jcz2",
+           "jliao2",
            # (9 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-sms-surface will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface" {
        id       = "scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43"
!       members  = [
+           "jcz2",
+           "jliao2",
            # (9 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-surface will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-surface" {
        id       = "scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775"
!       members  = [
+           "jcz2",
+           "jliao2",
            # (9 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "jcz2",
+           "jliao2",
            # (77 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 5 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 5 to change, 0 to destroy.


4. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.22s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # forgejo_team_member.slai_jcz2 will be created
+ resource "forgejo_team_member" "slai_jcz2" {
+       team_id = 8
+       user    = "jcz2"
    }

  # forgejo_team_member.slai_jliao2 will be created
+ resource "forgejo_team_member" "slai_jliao2" {
+       team_id = 8
+       user    = "jliao2"
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 0 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.09s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.jcz2 will be created
+ resource "discord_member_roles" "jcz2" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "844604301573750784"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1505078357321191535"
        }
    }

  # discord_member_roles.jliao2 will be created
+ resource "discord_member_roles" "jliao2" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "709410097536958624"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1505078357321191535"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 2 to add, 0 to change, 0 to destroy.


Plan Summary

5 projects, 5 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 5 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.20s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_hub_jcz2 will be created + resource "null_resource" "slack_hub_jcz2" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0BTHUYABJP" } } # null_resource.slack_hub_jliao2 will be created + resource "null_resource" "slack_hub_jliao2" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U09F23EKK6X" } } # null_resource.slack_slai_ch0_jcz2 will be created + resource "null_resource" "slack_slai_ch0_jcz2" { + id = (known after apply) + triggers = { + "channel" = "C08SRUGUQ05" + "user" = "U0BTHUYABJP" } } # null_resource.slack_slai_ch0_jliao2 will be created + resource "null_resource" "slack_slai_ch0_jliao2" { + id = (known after apply) + triggers = { + "channel" = "C08SRUGUQ05" + "user" = "U09F23EKK6X" } } Plan: 4 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 4 to add, 0 to change, 0 to destroy. --- ### 2. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.20s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_jcz2 will be created + resource "google_cloud_identity_group_membership" "tech_jcz2" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "jcz2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } # google_cloud_identity_group_membership.tech_jliao2 will be created + resource "google_cloud_identity_group_membership" "tech_jliao2" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "jliao2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 3. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.16s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_slai_cmugpt-agent will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-agent" { id = "scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423" ! members = [ + "jcz2", + "jliao2", # (9 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-mcp-server will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server" { id = "scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8" ! members = [ + "jcz2", + "jliao2", # (9 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-sms-surface will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface" { id = "scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43" ! members = [ + "jcz2", + "jliao2", # (9 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-surface will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-surface" { id = "scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775" ! members = [ + "jcz2", + "jliao2", # (9 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "jcz2", + "jliao2", # (77 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 5 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 5 to change, 0 to destroy. --- ### 4. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.22s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # forgejo_team_member.slai_jcz2 will be created + resource "forgejo_team_member" "slai_jcz2" { + team_id = 8 + user = "jcz2" } # forgejo_team_member.slai_jliao2 will be created + resource "forgejo_team_member" "slai_jliao2" { + team_id = 8 + user = "jliao2" } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.09s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.jcz2 will be created + resource "discord_member_roles" "jcz2" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "844604301573750784" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1505078357321191535" } } # discord_member_roles.jliao2 will be created + resource "discord_member_roles" "jliao2" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "709410097536958624" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1505078357321191535" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### Plan Summary 5 projects, 5 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis apply

atlantis apply

Ran Apply for 5 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: git dir: tofu/git workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: discord dir: tofu/discord workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.15s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_hub_jliao2: Creating...
null_resource.slack_hub_jliao2: Provisioning with 'local-exec'...
null_resource.slack_hub_jliao2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U09F23EKK6X"]
null_resource.slack_slai_ch0_jliao2: Creating...
null_resource.slack_slai_ch0_jcz2: Creating...
null_resource.slack_hub_jcz2: Creating...
null_resource.slack_slai_ch0_jliao2: Provisioning with 'local-exec'...
null_resource.slack_slai_ch0_jliao2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08SRUGUQ05 --user U09F23EKK6X"]
null_resource.slack_slai_ch0_jcz2: Provisioning with 'local-exec'...
null_resource.slack_slai_ch0_jcz2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08SRUGUQ05 --user U0BTHUYABJP"]
null_resource.slack_hub_jcz2: Provisioning with 'local-exec'...
null_resource.slack_hub_jcz2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0BTHUYABJP"]
null_resource.slack_slai_ch0_jcz2: Creation complete after 1s [id=8690310981055433365]
null_resource.slack_hub_jcz2: Creation complete after 1s [id=4843697392163247890]
null_resource.slack_slai_ch0_jliao2: Creation complete after 1s [id=3989805216577422639]
null_resource.slack_hub_jliao2: Creation complete after 1s [id=880534633743601555]

Apply complete! Resources: 4 added, 0 changed, 0 destroyed.

2. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.19s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.slai_jliao2: Creating...
forgejo_team_member.slai_jcz2: Creating...
forgejo_team_member.slai_jliao2: Creation complete after 0s
forgejo_team_member.slai_jcz2: Creation complete after 0s

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.16s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_jcz2: Creating...
google_cloud_identity_group_membership.tech_jliao2: Creating...
google_cloud_identity_group_membership.tech_jliao2: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/118411272773490859224]
google_cloud_identity_group_membership.tech_jcz2: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/108656568399472516729]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

4. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.jliao2: Creating...
discord_member_roles.jcz2: Creating...
discord_member_roles.jliao2: Creation complete after 2s [id=1353118209024331948:709410097536958624]
discord_member_roles.jcz2: Creation complete after 2s [id=1353118209024331948:844604301573750784]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

5. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.14s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_slai_cmugpt-mcp-server: Modifying... [id=scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8]
keycloak_group_memberships.project_slai_cmugpt-surface: Modifying... [id=scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775]
keycloak_group_memberships.project_slai_cmugpt-sms-surface: Modifying... [id=scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43]
keycloak_group_memberships.project_slai_cmugpt-agent: Modifying... [id=scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_slai_cmugpt-mcp-server: Modifications complete after 4s [id=scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8]
keycloak_group_memberships.project_slai_cmugpt-agent: Modifications complete after 4s [id=scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423]
keycloak_group_memberships.project_slai_cmugpt-surface: Modifications complete after 4s [id=scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775]
keycloak_group_memberships.project_slai_cmugpt-sms-surface: Modifications complete after 4s [id=scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43]
keycloak_group_memberships.projects: Modifications complete after 5s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 5 changed, 0 destroyed.

Apply Summary

5 projects, 5 successful, 0 failed, 0 errored

Ran Apply for 5 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.15s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_hub_jliao2: Creating... null_resource.slack_hub_jliao2: Provisioning with 'local-exec'... null_resource.slack_hub_jliao2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U09F23EKK6X"] null_resource.slack_slai_ch0_jliao2: Creating... null_resource.slack_slai_ch0_jcz2: Creating... null_resource.slack_hub_jcz2: Creating... null_resource.slack_slai_ch0_jliao2: Provisioning with 'local-exec'... null_resource.slack_slai_ch0_jliao2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08SRUGUQ05 --user U09F23EKK6X"] null_resource.slack_slai_ch0_jcz2: Provisioning with 'local-exec'... null_resource.slack_slai_ch0_jcz2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08SRUGUQ05 --user U0BTHUYABJP"] null_resource.slack_hub_jcz2: Provisioning with 'local-exec'... null_resource.slack_hub_jcz2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0BTHUYABJP"] null_resource.slack_slai_ch0_jcz2: Creation complete after 1s [id=8690310981055433365] null_resource.slack_hub_jcz2: Creation complete after 1s [id=4843697392163247890] null_resource.slack_slai_ch0_jliao2: Creation complete after 1s [id=3989805216577422639] null_resource.slack_hub_jliao2: Creation complete after 1s [id=880534633743601555] Apply complete! Resources: 4 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.19s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.slai_jliao2: Creating... forgejo_team_member.slai_jcz2: Creating... forgejo_team_member.slai_jliao2: Creation complete after 0s forgejo_team_member.slai_jcz2: Creation complete after 0s Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.16s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_jcz2: Creating... google_cloud_identity_group_membership.tech_jliao2: Creating... google_cloud_identity_group_membership.tech_jliao2: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/118411272773490859224] google_cloud_identity_group_membership.tech_jcz2: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/108656568399472516729] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.jliao2: Creating... discord_member_roles.jcz2: Creating... discord_member_roles.jliao2: Creation complete after 2s [id=1353118209024331948:709410097536958624] discord_member_roles.jcz2: Creation complete after 2s [id=1353118209024331948:844604301573750784] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.14s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_slai_cmugpt-mcp-server: Modifying... [id=scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8] keycloak_group_memberships.project_slai_cmugpt-surface: Modifying... [id=scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775] keycloak_group_memberships.project_slai_cmugpt-sms-surface: Modifying... [id=scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43] keycloak_group_memberships.project_slai_cmugpt-agent: Modifying... [id=scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_slai_cmugpt-mcp-server: Modifications complete after 4s [id=scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8] keycloak_group_memberships.project_slai_cmugpt-agent: Modifications complete after 4s [id=scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423] keycloak_group_memberships.project_slai_cmugpt-surface: Modifications complete after 4s [id=scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775] keycloak_group_memberships.project_slai_cmugpt-sms-surface: Modifications complete after 4s [id=scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43] keycloak_group_memberships.projects: Modifications complete after 5s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 5 changed, 0 destroyed. ``` </details> --- ### Apply Summary 5 projects, 5 successful, 0 failed, 0 errored
anish merged commit 6ea64a3b6c into main 2026-09-22 12:41:00 +00:00
anish deleted branch add-slai-unapproved 2026-09-22 12:41:00 +00:00

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
Sign in to join this conversation.
No description provided.