feat: add akhanal2 to slai #301

Merged
xboxbedrock merged 2 commits from akhanal2/governance:akhanal2-patch-1 into main 2026-09-20 11:01:07 +00:00
Member

Added akhanal2 to slai.toml according to governence rules

Added akhanal2 to slai.toml according to governence rules
feat: add akhanal2 to slai
Some checks are pending
CI / check (pull_request) Blocked by required conditions
CI / check-1 (pull_request) Blocked by required conditions
CI / build (pull_request) Blocked by required conditions
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 17m32s
85f55516a1
Added akhanal2 to slai.toml according to governence rules
Author
Member

Proof of linkage(only one missing is the optional Codeberg)

Proof of linkage(only one missing is the optional Codeberg)
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: keycloak Plan failed.
atlantis/plan: slack Plan: 2 to add, 0 to change, 2 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 2 to destroy.
atlantis/plan: discord Plan: 1 to add, 2 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
atlantis/plan: git Plan: 2 to add, 2 to change, 10 to destroy.
atlantis/plan 1/6 projects failed to plan.
1902bd7bf2

Ran Plan for 6 projects:

  1. project: keycloak dir: tofu/keycloak workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: google_play dir: tofu/google_play workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #267. To continue, delete the lock from #267 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 8.57s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_hub_akhanal2 will be created
+ resource "null_resource" "slack_hub_akhanal2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0C201A8QQ3"
        }
    }

  # null_resource.slack_leads_jc_965 will be destroyed
  # (because null_resource.slack_leads_jc_965 is not in configuration)
- resource "null_resource" "slack_leads_jc_965" {
-       id       = "6185492700614965496" -> null
-       triggers = {
          - "channel" = "C0AN1M0HDV0"
          - "user"    = "U09E19AABEY"
        } -> null
    }

  # null_resource.slack_leads_jiyas will be destroyed
  # (because null_resource.slack_leads_jiyas is not in configuration)
- resource "null_resource" "slack_leads_jiyas" {
-       id       = "1866307958230487938" -> null
-       triggers = {
          - "channel" = "C0AN1M0HDV0"
          - "user"    = "U09F6FZH78A"
        } -> null
    }

  # null_resource.slack_slai_ch0_akhanal2 will be created
+ resource "null_resource" "slack_slai_ch0_akhanal2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08SRUGUQ05"
          + "user"    = "U0C201A8QQ3"
        }
    }

Plan: 2 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 2 to destroy.


3. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
   Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/301/default/crates/governance-schema)
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/301/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/301/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/301/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 8.59s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

No changes. Your infrastructure matches the configuration.


4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 8.60s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.ops_jc_965 will be destroyed
  # (because google_cloud_identity_group_membership.ops_jc_965 is not in configuration)
- resource "google_cloud_identity_group_membership" "ops_jc_965" {
-       create_time = "2026-09-20T00:28:09.056393Z" -> null
-       group       = "groups/03fwokq036zsp1f" -> null
-       id          = "groups/03fwokq036zsp1f/memberships/102552619255005499474" -> null
-       name        = "groups/03fwokq036zsp1f/memberships/102552619255005499474" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-20T00:28:09.056393Z" -> null

-       preferred_member_key {
-           id = "jessec2@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.ops_jiyas will be destroyed
  # (because google_cloud_identity_group_membership.ops_jiyas is not in configuration)
- resource "google_cloud_identity_group_membership" "ops_jiyas" {
-       create_time = "2026-09-20T00:28:09.052968Z" -> null
-       group       = "groups/03fwokq036zsp1f" -> null
-       id          = "groups/03fwokq036zsp1f/memberships/116261093135461470038" -> null
-       name        = "groups/03fwokq036zsp1f/memberships/116261093135461470038" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-20T00:28:09.052968Z" -> null

-       preferred_member_key {
-           id = "jiyas@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.tech_akhanal2 will be created
+ resource "google_cloud_identity_group_membership" "tech_akhanal2" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "akhanal2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 2 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 8.60s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # discord_member_roles.akhanal2 will be created
+ resource "discord_member_roles" "akhanal2" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "640268047994716180"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1505078357321191535"
        }
    }

  # discord_member_roles.jc_965 will be updated in-place
! resource "discord_member_roles" "jc_965" {
        id        = "1353118209024331948:472617578938761226"
        # (2 unchanged attributes hidden)

-       role {
-           has_role = true -> null
-           role_id  = "1519792320755662928" -> null
        }

        # (2 unchanged blocks hidden)
    }

  # discord_member_roles.jiyas will be updated in-place
! resource "discord_member_roles" "jiyas" {
        id        = "1353118209024331948:829775805768925275"
        # (2 unchanged attributes hidden)

-       role {
-           has_role = true -> null
-           role_id  = "1519792320755662928" -> null
        }

        # (2 unchanged blocks hidden)
    }

Plan: 1 to add, 2 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 2 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 8.60s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)
- destroy

OpenTofu will perform the following actions:

  # forgejo_team_member.cmugpt-agent_leads_jc_965 will be destroyed
  # (because forgejo_team_member.cmugpt-agent_leads_jc_965 is not in configuration)
- resource "forgejo_team_member" "cmugpt-agent_leads_jc_965" {
-       team_id = 24 -> null
-       user    = "jessec2" -> null
    }

  # forgejo_team_member.cmugpt-agent_leads_jiyas will be destroyed
  # (because forgejo_team_member.cmugpt-agent_leads_jiyas is not in configuration)
- resource "forgejo_team_member" "cmugpt-agent_leads_jiyas" {
-       team_id = 24 -> null
-       user    = "jiyas" -> null
    }

  # forgejo_team_member.cmugpt-mcp-server_leads_jc_965 will be destroyed
  # (because forgejo_team_member.cmugpt-mcp-server_leads_jc_965 is not in configuration)
- resource "forgejo_team_member" "cmugpt-mcp-server_leads_jc_965" {
-       team_id = 26 -> null
-       user    = "jessec2" -> null
    }

  # forgejo_team_member.cmugpt-mcp-server_leads_jiyas will be destroyed
  # (because forgejo_team_member.cmugpt-mcp-server_leads_jiyas is not in configuration)
- resource "forgejo_team_member" "cmugpt-mcp-server_leads_jiyas" {
-       team_id = 26 -> null
-       user    = "jiyas" -> null
    }

  # forgejo_team_member.cmugpt-sms-surface_leads_jc_965 will be destroyed
  # (because forgejo_team_member.cmugpt-sms-surface_leads_jc_965 is not in configuration)
- resource "forgejo_team_member" "cmugpt-sms-surface_leads_jc_965" {
-       team_id = 25 -> null
-       user    = "jessec2" -> null
    }

  # forgejo_team_member.cmugpt-sms-surface_leads_jiyas will be destroyed
  # (because forgejo_team_member.cmugpt-sms-surface_leads_jiyas is not in configuration)
- resource "forgejo_team_member" "cmugpt-sms-surface_leads_jiyas" {
-       team_id = 25 -> null
-       user    = "jiyas" -> null
    }

  # forgejo_team_member.cmugpt-surface_leads_jc_965 will be destroyed
  # (because forgejo_team_member.cmugpt-surface_leads_jc_965 is not in configuration)
- resource "forgejo_team_member" "cmugpt-surface_leads_jc_965" {
-       team_id = 23 -> null
-       user    = "jessec2" -> null
    }

  # forgejo_team_member.cmugpt-surface_leads_jiyas will be destroyed
  # (because forgejo_team_member.cmugpt-surface_leads_jiyas is not in configuration)
- resource "forgejo_team_member" "cmugpt-surface_leads_jiyas" {
-       team_id = 23 -> null
-       user    = "jiyas" -> null
    }

  # forgejo_team_member.slai_akhanal2 will be created
+ resource "forgejo_team_member" "slai_akhanal2" {
+       team_id = 8
+       user    = "akhanal2"
    }

  # forgejo_team_member.tech_leads_jc_965 will be destroyed
  # (because forgejo_team_member.tech_leads_jc_965 is not in configuration)
- resource "forgejo_team_member" "tech_leads_jc_965" {
-       team_id = 44 -> null
-       user    = "jessec2" -> null
    }

  # forgejo_team_member.tech_leads_jiyas will be destroyed
  # (because forgejo_team_member.tech_leads_jiyas is not in configuration)
- resource "forgejo_team_member" "tech_leads_jiyas" {
-       team_id = 44 -> null
-       user    = "jiyas" -> null
    }

  # github_team_membership.slai_akhanal2 will be created
+ resource "github_team_membership" "slai_akhanal2" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "17571328"
+       username = "okak1-pwn"
    }

  # github_team_membership.slai_jc_965 will be updated in-place
! resource "github_team_membership" "slai_jc_965" {
        id       = "17571328:jc-965"
!       role     = "maintainer" -> "member"
        # (3 unchanged attributes hidden)
    }

  # github_team_membership.slai_jiyas will be updated in-place
! resource "github_team_membership" "slai_jiyas" {
        id       = "17571328:jiya2301"
!       role     = "maintainer" -> "member"
        # (3 unchanged attributes hidden)
    }

Plan: 2 to add, 2 to change, 10 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 2 to change, 10 to destroy.


Plan Summary

6 projects, 4 with changes, 1 with no changes, 1 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #267. To continue, delete the lock from #267 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 8.57s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # null_resource.slack_hub_akhanal2 will be created + resource "null_resource" "slack_hub_akhanal2" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0C201A8QQ3" } } # null_resource.slack_leads_jc_965 will be destroyed # (because null_resource.slack_leads_jc_965 is not in configuration) - resource "null_resource" "slack_leads_jc_965" { - id = "6185492700614965496" -> null - triggers = { - "channel" = "C0AN1M0HDV0" - "user" = "U09E19AABEY" } -> null } # null_resource.slack_leads_jiyas will be destroyed # (because null_resource.slack_leads_jiyas is not in configuration) - resource "null_resource" "slack_leads_jiyas" { - id = "1866307958230487938" -> null - triggers = { - "channel" = "C0AN1M0HDV0" - "user" = "U09F6FZH78A" } -> null } # null_resource.slack_slai_ch0_akhanal2 will be created + resource "null_resource" "slack_slai_ch0_akhanal2" { + id = (known after apply) + triggers = { + "channel" = "C08SRUGUQ05" + "user" = "U0C201A8QQ3" } } Plan: 2 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 2 to destroy. --- ### 3. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/301/default/crates/governance-schema) Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/301/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/301/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/301/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 8.59s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` No changes. Your infrastructure matches the configuration. --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 8.60s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # google_cloud_identity_group_membership.ops_jc_965 will be destroyed # (because google_cloud_identity_group_membership.ops_jc_965 is not in configuration) - resource "google_cloud_identity_group_membership" "ops_jc_965" { - create_time = "2026-09-20T00:28:09.056393Z" -> null - group = "groups/03fwokq036zsp1f" -> null - id = "groups/03fwokq036zsp1f/memberships/102552619255005499474" -> null - name = "groups/03fwokq036zsp1f/memberships/102552619255005499474" -> null - type = "USER" -> null - update_time = "2026-09-20T00:28:09.056393Z" -> null - preferred_member_key { - id = "jessec2@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.ops_jiyas will be destroyed # (because google_cloud_identity_group_membership.ops_jiyas is not in configuration) - resource "google_cloud_identity_group_membership" "ops_jiyas" { - create_time = "2026-09-20T00:28:09.052968Z" -> null - group = "groups/03fwokq036zsp1f" -> null - id = "groups/03fwokq036zsp1f/memberships/116261093135461470038" -> null - name = "groups/03fwokq036zsp1f/memberships/116261093135461470038" -> null - type = "USER" -> null - update_time = "2026-09-20T00:28:09.052968Z" -> null - preferred_member_key { - id = "jiyas@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.tech_akhanal2 will be created + resource "google_cloud_identity_group_membership" "tech_akhanal2" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "akhanal2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 2 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 8.60s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # discord_member_roles.akhanal2 will be created + resource "discord_member_roles" "akhanal2" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "640268047994716180" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1505078357321191535" } } # discord_member_roles.jc_965 will be updated in-place ! resource "discord_member_roles" "jc_965" { id = "1353118209024331948:472617578938761226" # (2 unchanged attributes hidden) - role { - has_role = true -> null - role_id = "1519792320755662928" -> null } # (2 unchanged blocks hidden) } # discord_member_roles.jiyas will be updated in-place ! resource "discord_member_roles" "jiyas" { id = "1353118209024331948:829775805768925275" # (2 unchanged attributes hidden) - role { - has_role = true -> null - role_id = "1519792320755662928" -> null } # (2 unchanged blocks hidden) } Plan: 1 to add, 2 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 2 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 8.60s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) - destroy OpenTofu will perform the following actions: # forgejo_team_member.cmugpt-agent_leads_jc_965 will be destroyed # (because forgejo_team_member.cmugpt-agent_leads_jc_965 is not in configuration) - resource "forgejo_team_member" "cmugpt-agent_leads_jc_965" { - team_id = 24 -> null - user = "jessec2" -> null } # forgejo_team_member.cmugpt-agent_leads_jiyas will be destroyed # (because forgejo_team_member.cmugpt-agent_leads_jiyas is not in configuration) - resource "forgejo_team_member" "cmugpt-agent_leads_jiyas" { - team_id = 24 -> null - user = "jiyas" -> null } # forgejo_team_member.cmugpt-mcp-server_leads_jc_965 will be destroyed # (because forgejo_team_member.cmugpt-mcp-server_leads_jc_965 is not in configuration) - resource "forgejo_team_member" "cmugpt-mcp-server_leads_jc_965" { - team_id = 26 -> null - user = "jessec2" -> null } # forgejo_team_member.cmugpt-mcp-server_leads_jiyas will be destroyed # (because forgejo_team_member.cmugpt-mcp-server_leads_jiyas is not in configuration) - resource "forgejo_team_member" "cmugpt-mcp-server_leads_jiyas" { - team_id = 26 -> null - user = "jiyas" -> null } # forgejo_team_member.cmugpt-sms-surface_leads_jc_965 will be destroyed # (because forgejo_team_member.cmugpt-sms-surface_leads_jc_965 is not in configuration) - resource "forgejo_team_member" "cmugpt-sms-surface_leads_jc_965" { - team_id = 25 -> null - user = "jessec2" -> null } # forgejo_team_member.cmugpt-sms-surface_leads_jiyas will be destroyed # (because forgejo_team_member.cmugpt-sms-surface_leads_jiyas is not in configuration) - resource "forgejo_team_member" "cmugpt-sms-surface_leads_jiyas" { - team_id = 25 -> null - user = "jiyas" -> null } # forgejo_team_member.cmugpt-surface_leads_jc_965 will be destroyed # (because forgejo_team_member.cmugpt-surface_leads_jc_965 is not in configuration) - resource "forgejo_team_member" "cmugpt-surface_leads_jc_965" { - team_id = 23 -> null - user = "jessec2" -> null } # forgejo_team_member.cmugpt-surface_leads_jiyas will be destroyed # (because forgejo_team_member.cmugpt-surface_leads_jiyas is not in configuration) - resource "forgejo_team_member" "cmugpt-surface_leads_jiyas" { - team_id = 23 -> null - user = "jiyas" -> null } # forgejo_team_member.slai_akhanal2 will be created + resource "forgejo_team_member" "slai_akhanal2" { + team_id = 8 + user = "akhanal2" } # forgejo_team_member.tech_leads_jc_965 will be destroyed # (because forgejo_team_member.tech_leads_jc_965 is not in configuration) - resource "forgejo_team_member" "tech_leads_jc_965" { - team_id = 44 -> null - user = "jessec2" -> null } # forgejo_team_member.tech_leads_jiyas will be destroyed # (because forgejo_team_member.tech_leads_jiyas is not in configuration) - resource "forgejo_team_member" "tech_leads_jiyas" { - team_id = 44 -> null - user = "jiyas" -> null } # github_team_membership.slai_akhanal2 will be created + resource "github_team_membership" "slai_akhanal2" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "17571328" + username = "okak1-pwn" } # github_team_membership.slai_jc_965 will be updated in-place ! resource "github_team_membership" "slai_jc_965" { id = "17571328:jc-965" ! role = "maintainer" -> "member" # (3 unchanged attributes hidden) } # github_team_membership.slai_jiyas will be updated in-place ! resource "github_team_membership" "slai_jiyas" { id = "17571328:jiya2301" ! role = "maintainer" -> "member" # (3 unchanged attributes hidden) } Plan: 2 to add, 2 to change, 10 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 2 to change, 10 to destroy. --- ### Plan Summary 6 projects, 4 with changes, 1 with no changes, 1 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
xboxbedrock force-pushed akhanal2-patch-1 from 1902bd7bf2
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: keycloak Plan failed.
atlantis/plan: slack Plan: 2 to add, 0 to change, 2 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 2 to destroy.
atlantis/plan: discord Plan: 1 to add, 2 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
atlantis/plan: git Plan: 2 to add, 2 to change, 10 to destroy.
atlantis/plan 1/6 projects failed to plan.
to a95ee63871
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 5 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 2 to add, 1 to change, 0 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
2026-09-20 10:56:51 +00:00
Compare

Ran Plan for 6 projects:

  1. project: keycloak dir: tofu/keycloak workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: google_play dir: tofu/google_play workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #267. To continue, delete the lock from #267 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_hub_akhanal2 will be created
+ resource "null_resource" "slack_hub_akhanal2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0C201A8QQ3"
        }
    }

  # null_resource.slack_slai_ch0_akhanal2 will be created
+ resource "null_resource" "slack_slai_ch0_akhanal2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08SRUGUQ05"
          + "user"    = "U0C201A8QQ3"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 0 to destroy.


3. project: google_play dir: tofu/google_play workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.15s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.19s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_akhanal2 will be created
+ resource "google_cloud_identity_group_membership" "tech_akhanal2" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "akhanal2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.21s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.akhanal2 will be created
+ resource "discord_member_roles" "akhanal2" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "640268047994716180"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1505078357321191535"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.20s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # forgejo_team_member.slai_akhanal2 will be created
+ resource "forgejo_team_member" "slai_akhanal2" {
+       team_id = 8
+       user    = "akhanal2"
    }

  # github_repository.slai_cmugpt_agent will be updated in-place
! resource "github_repository" "slai_cmugpt_agent" {
        id                                      = "cmugpt-agent"
        name                                    = "cmugpt-agent"
        # (39 unchanged attributes hidden)

-       template {
-           include_all_branches = false -> null
-           owner                = "ScottyLabs" -> null
-           repository           = "python-template" -> null
        }

        # (1 unchanged block hidden)
    }

  # github_team_membership.slai_akhanal2 will be created
+ resource "github_team_membership" "slai_akhanal2" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "17571328"
+       username = "okak1-pwn"
    }

Plan: 2 to add, 1 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 1 to change, 0 to destroy.


Plan Summary

6 projects, 4 with changes, 1 with no changes, 1 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #267. To continue, delete the lock from #267 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_hub_akhanal2 will be created + resource "null_resource" "slack_hub_akhanal2" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0C201A8QQ3" } } # null_resource.slack_slai_ch0_akhanal2 will be created + resource "null_resource" "slack_slai_ch0_akhanal2" { + id = (known after apply) + triggers = { + "channel" = "C08SRUGUQ05" + "user" = "U0C201A8QQ3" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 3. project: `google_play` dir: `tofu/google_play` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.15s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.19s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_akhanal2 will be created + resource "google_cloud_identity_group_membership" "tech_akhanal2" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "akhanal2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.21s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.akhanal2 will be created + resource "discord_member_roles" "akhanal2" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "640268047994716180" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1505078357321191535" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.20s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # forgejo_team_member.slai_akhanal2 will be created + resource "forgejo_team_member" "slai_akhanal2" { + team_id = 8 + user = "akhanal2" } # github_repository.slai_cmugpt_agent will be updated in-place ! resource "github_repository" "slai_cmugpt_agent" { id = "cmugpt-agent" name = "cmugpt-agent" # (39 unchanged attributes hidden) - template { - include_all_branches = false -> null - owner = "ScottyLabs" -> null - repository = "python-template" -> null } # (1 unchanged block hidden) } # github_team_membership.slai_akhanal2 will be created + resource "github_team_membership" "slai_akhanal2" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "17571328" + username = "okak1-pwn" } Plan: 2 to add, 1 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 1 to change, 0 to destroy. --- ### Plan Summary 6 projects, 4 with changes, 1 with no changes, 1 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.20s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_hub_akhanal2 will be created
+ resource "null_resource" "slack_hub_akhanal2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0C201A8QQ3"
        }
    }

  # null_resource.slack_slai_ch0_akhanal2 will be created
+ resource "null_resource" "slack_slai_ch0_akhanal2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08SRUGUQ05"
          + "user"    = "U0C201A8QQ3"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 0 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.20s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.16s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_akhanal2 will be created
+ resource "google_cloud_identity_group_membership" "tech_akhanal2" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "akhanal2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.21s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_slai_cmugpt-agent will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-agent" {
        id       = "scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423"
!       members  = [
+           "akhanal2",
            # (7 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-mcp-server will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server" {
        id       = "scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8"
!       members  = [
+           "akhanal2",
            # (7 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-sms-surface will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface" {
        id       = "scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43"
!       members  = [
+           "akhanal2",
            # (7 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-surface will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-surface" {
        id       = "scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775"
!       members  = [
+           "akhanal2",
            # (7 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "akhanal2",
            # (50 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 5 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 5 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.20s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.akhanal2 will be created
+ resource "discord_member_roles" "akhanal2" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "640268047994716180"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1505078357321191535"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.19s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # forgejo_team_member.slai_akhanal2 will be created
+ resource "forgejo_team_member" "slai_akhanal2" {
+       team_id = 8
+       user    = "akhanal2"
    }

  # github_repository.slai_cmugpt_agent will be updated in-place
! resource "github_repository" "slai_cmugpt_agent" {
        id                                      = "cmugpt-agent"
        name                                    = "cmugpt-agent"
        # (39 unchanged attributes hidden)

-       template {
-           include_all_branches = false -> null
-           owner                = "ScottyLabs" -> null
-           repository           = "python-template" -> null
        }

        # (1 unchanged block hidden)
    }

  # github_team_membership.slai_akhanal2 will be created
+ resource "github_team_membership" "slai_akhanal2" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "17571328"
+       username = "okak1-pwn"
    }

Plan: 2 to add, 1 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 1 to change, 0 to destroy.


Plan Summary

6 projects, 5 with changes, 1 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.20s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_hub_akhanal2 will be created + resource "null_resource" "slack_hub_akhanal2" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0C201A8QQ3" } } # null_resource.slack_slai_ch0_akhanal2 will be created + resource "null_resource" "slack_slai_ch0_akhanal2" { + id = (known after apply) + triggers = { + "channel" = "C08SRUGUQ05" + "user" = "U0C201A8QQ3" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.20s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.16s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_akhanal2 will be created + resource "google_cloud_identity_group_membership" "tech_akhanal2" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "akhanal2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.21s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_slai_cmugpt-agent will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-agent" { id = "scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423" ! members = [ + "akhanal2", # (7 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-mcp-server will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server" { id = "scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8" ! members = [ + "akhanal2", # (7 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-sms-surface will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface" { id = "scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43" ! members = [ + "akhanal2", # (7 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-surface will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-surface" { id = "scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775" ! members = [ + "akhanal2", # (7 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "akhanal2", # (50 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 5 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 5 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.20s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.akhanal2 will be created + resource "discord_member_roles" "akhanal2" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "640268047994716180" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1505078357321191535" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.19s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # forgejo_team_member.slai_akhanal2 will be created + resource "forgejo_team_member" "slai_akhanal2" { + team_id = 8 + user = "akhanal2" } # github_repository.slai_cmugpt_agent will be updated in-place ! resource "github_repository" "slai_cmugpt_agent" { id = "cmugpt-agent" name = "cmugpt-agent" # (39 unchanged attributes hidden) - template { - include_all_branches = false -> null - owner = "ScottyLabs" -> null - repository = "python-template" -> null } # (1 unchanged block hidden) } # github_team_membership.slai_akhanal2 will be created + resource "github_team_membership" "slai_akhanal2" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "17571328" + username = "okak1-pwn" } Plan: 2 to add, 1 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 1 to change, 0 to destroy. --- ### Plan Summary 6 projects, 5 with changes, 1 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis apply

atlantis apply

Ran Apply for 6 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.11s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

2. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.16s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_hub_akhanal2: Creating...
null_resource.slack_slai_ch0_akhanal2: Creating...
null_resource.slack_hub_akhanal2: Provisioning with 'local-exec'...
null_resource.slack_hub_akhanal2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C201A8QQ3"]
null_resource.slack_slai_ch0_akhanal2: Provisioning with 'local-exec'...
null_resource.slack_slai_ch0_akhanal2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08SRUGUQ05 --user U0C201A8QQ3"]
null_resource.slack_hub_akhanal2: Creation complete after 1s [id=7051117878881064331]
null_resource.slack_slai_ch0_akhanal2: Creation complete after 1s [id=5858842237558273895]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

3. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.17s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.akhanal2: Creating...
discord_member_roles.akhanal2: Creation complete after 1s [id=1353118209024331948:640268047994716180]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.22s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_akhanal2: Creating...
google_cloud_identity_group_membership.tech_akhanal2: Creation complete after 0s [id=groups/04anzqyu1bqbv79/memberships/107313041332888979038]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

5. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_slai_cmugpt-surface: Modifying... [id=scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775]
keycloak_group_memberships.project_slai_cmugpt-mcp-server: Modifying... [id=scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8]
keycloak_group_memberships.project_slai_cmugpt-sms-surface: Modifying... [id=scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43]
keycloak_group_memberships.project_slai_cmugpt-agent: Modifying... [id=scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_slai_cmugpt-sms-surface: Modifications complete after 3s [id=scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43]
keycloak_group_memberships.project_slai_cmugpt-agent: Modifications complete after 3s [id=scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423]
keycloak_group_memberships.project_slai_cmugpt-surface: Modifications complete after 3s [id=scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775]
keycloak_group_memberships.project_slai_cmugpt-mcp-server: Modifications complete after 3s [id=scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8]
keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 5 changed, 0 destroyed.

6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.29s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
github_team_membership.slai_akhanal2: Creating...
github_repository.slai_cmugpt_agent: Modifying... [id=cmugpt-agent]
forgejo_team_member.slai_akhanal2: Creating...
forgejo_team_member.slai_akhanal2: Creation complete after 0s
github_team_membership.slai_akhanal2: Creation complete after 3s [id=17571328:okak1-pwn]
github_repository.slai_cmugpt_agent: Modifications complete after 4s [id=cmugpt-agent]

Apply complete! Resources: 2 added, 1 changed, 0 destroyed.

Apply Summary

6 projects, 6 successful, 0 failed, 0 errored

Ran Apply for 6 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.11s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.16s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_hub_akhanal2: Creating... null_resource.slack_slai_ch0_akhanal2: Creating... null_resource.slack_hub_akhanal2: Provisioning with 'local-exec'... null_resource.slack_hub_akhanal2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C201A8QQ3"] null_resource.slack_slai_ch0_akhanal2: Provisioning with 'local-exec'... null_resource.slack_slai_ch0_akhanal2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08SRUGUQ05 --user U0C201A8QQ3"] null_resource.slack_hub_akhanal2: Creation complete after 1s [id=7051117878881064331] null_resource.slack_slai_ch0_akhanal2: Creation complete after 1s [id=5858842237558273895] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.17s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.akhanal2: Creating... discord_member_roles.akhanal2: Creation complete after 1s [id=1353118209024331948:640268047994716180] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.22s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_akhanal2: Creating... google_cloud_identity_group_membership.tech_akhanal2: Creation complete after 0s [id=groups/04anzqyu1bqbv79/memberships/107313041332888979038] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_slai_cmugpt-surface: Modifying... [id=scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775] keycloak_group_memberships.project_slai_cmugpt-mcp-server: Modifying... [id=scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8] keycloak_group_memberships.project_slai_cmugpt-sms-surface: Modifying... [id=scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43] keycloak_group_memberships.project_slai_cmugpt-agent: Modifying... [id=scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_slai_cmugpt-sms-surface: Modifications complete after 3s [id=scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43] keycloak_group_memberships.project_slai_cmugpt-agent: Modifications complete after 3s [id=scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423] keycloak_group_memberships.project_slai_cmugpt-surface: Modifications complete after 3s [id=scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775] keycloak_group_memberships.project_slai_cmugpt-mcp-server: Modifications complete after 3s [id=scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8] keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 5 changed, 0 destroyed. ``` </details> --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.29s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ github_team_membership.slai_akhanal2: Creating... github_repository.slai_cmugpt_agent: Modifying... [id=cmugpt-agent] forgejo_team_member.slai_akhanal2: Creating... forgejo_team_member.slai_akhanal2: Creation complete after 0s github_team_membership.slai_akhanal2: Creation complete after 3s [id=17571328:okak1-pwn] github_repository.slai_cmugpt_agent: Modifications complete after 4s [id=cmugpt-agent] Apply complete! Resources: 2 added, 1 changed, 0 destroyed. ``` </details> --- ### Apply Summary 6 projects, 6 successful, 0 failed, 0 errored

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
xboxbedrock referenced this pull request from a commit 2026-09-20 11:01:08 +00:00
akhanal2 deleted branch akhanal2-patch-1 2026-09-21 20:17:09 +00:00
brightz referenced this pull request from a commit 2026-09-26 22:03:28 +00:00
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 5 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 2 to add, 1 to change, 0 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
Some workflows are waiting to be reviewed.
Sign in to join this conversation.
No description provided.