feat: add kalee2 to cbp #267

Merged
xboxbedrock merged 3 commits from kalee2/governance:main into main 2026-09-21 03:55:12 +00:00
Member
No description provided.
Added name to cbp member
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
Validate / Governance (pull_request_target) Successful in 9m43s
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
CI / build (pull_request) Has been cancelled
6999895990
kalee2 changed title from add kalee2 to cbp to feat : add kalee2 to cbp 2026-09-19 20:54:41 +00:00
kalee2 changed title from feat : add kalee2 to cbp to feat: add kalee2 to cbp 2026-09-19 20:55:07 +00:00
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: keycloak Plan failed.
atlantis/plan 1/1 projects failed to plan.
kennel/build skipped: no services or static sites defined
074d8cc061

Ran Plan for dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #266. To continue, delete the lock from #266 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.

Ran Plan for dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #266. To continue, delete the lock from #266 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan.
tentype requested changes 2026-09-20 00:05:44 +00:00
Dismissed
@ -10,6 +10,7 @@ members = [
# "zeyuyaoy",
# "stationedk-06",
# "jay3332",
# "kalee2",
Member

This line is commented so it doesn't have any effect. Could you remove the #?

It should be

    "kalee2",

Thanks!

This line is commented so it doesn't have any effect. Could you remove the `#`? It should be ``` "kalee2", ``` Thanks!
kalee2 marked this conversation as resolved
fix: uncomment kalee2 in team members list
Some checks are pending
CI / check (pull_request) Blocked by required conditions
CI / check-1 (pull_request) Blocked by required conditions
CI / build (pull_request) Blocked by required conditions
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: keycloak Plan: 0 to add, 19 to change, 0 to destroy.
atlantis/plan 1/1 projects planned successfully.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 14m23s
bf5578ee0c

Ran Plan for project: keycloak dir: tofu/keycloak workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.07s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cbp_bus-sign will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_bus-sign" {
        id       = "scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b"
!       members  = [
-           "joohank",
            # (6 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_dalmatian will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_dalmatian" {
        id       = "scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153"
!       members  = [
-           "joohank",
            # (6 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_discord-verify will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_discord-verify" {
        id       = "scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745"
!       members  = [
-           "joohank",
            # (6 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_groupme-mirror will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_groupme-mirror" {
        id       = "scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39"
!       members  = [
-           "joohank",
            # (6 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-courses will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-courses" {
        id       = "scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77"
!       members  = [
-           "ckawibow",
-           "kyma",
-           "sgli",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-courses_courses will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-courses_courses" {
        id       = "scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406"
!       members  = [
-           "ckawibow",
-           "kyma",
-           "sgli",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-housing_housing" {
        id       = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d"
!       members  = [
-           "xindiy",
            # (13 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-maps_maps will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-maps_maps" {
        id       = "scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64"
!       members  = [
-           "arnavrao",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-study_study will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-study_study" {
        id       = "scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a"
!       members  = [
-           "andyj",
-           "dsoral",
-           "klaipras",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops will be updated in-place
! resource "keycloak_group_memberships" "project_devops" {
        id       = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7"
!       members  = [
-           "joohank",
-           "rjwu",
-           "saniyav",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_changelog will be updated in-place
! resource "keycloak_group_memberships" "project_devops_changelog" {
        id       = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77"
!       members  = [
-           "joohank",
-           "rjwu",
-           "saniyav",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_documentation will be updated in-place
! resource "keycloak_group_memberships" "project_devops_documentation" {
        id       = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f"
!       members  = [
-           "joohank",
-           "rjwu",
-           "saniyav",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_link-shortener will be updated in-place
! resource "keycloak_group_memberships" "project_devops_link-shortener" {
        id       = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935"
!       members  = [
-           "joohank",
-           "rjwu",
-           "saniyav",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-agent_admins will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-agent_admins" {
        id       = "scottylabs/group-memberships/f4e82f9b-769d-4f38-a7f2-e214189d9269"
!       members  = [
-           "jessec2",
-           "jiyas",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-mcp-server_admins will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server_admins" {
        id       = "scottylabs/group-memberships/8f84f9ec-e78a-41f6-9ce6-097fce273298"
!       members  = [
-           "jessec2",
-           "jiyas",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-sms-surface_admins will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface_admins" {
        id       = "scottylabs/group-memberships/34404457-985f-4a61-b2f8-4754c2eb3293"
!       members  = [
-           "jessec2",
-           "jiyas",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-surface_admins will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-surface_admins" {
        id       = "scottylabs/group-memberships/4f3531db-8075-480e-bbf4-253e98783b56"
!       members  = [
-           "jessec2",
-           "jiyas",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_tartan-vote will be updated in-place
! resource "keycloak_group_memberships" "project_tartan-vote" {
        id       = "scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c"
!       members  = [
-           "drishtis",
-           "joohank",
-           "jzd",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
-           "andyj",
-           "arnavrao",
-           "ckawibow",
-           "drishtis",
-           "dsoral",
-           "joohank",
-           "jzd",
-           "klaipras",
-           "kyma",
-           "rjwu",
-           "saniyav",
-           "sgli",
-           "xindiy",
            # (37 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 19 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 19 to change, 0 to destroy.


  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.07s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cbp_bus-sign will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_bus-sign" { id = "scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b" ! members = [ - "joohank", # (6 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_dalmatian will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_dalmatian" { id = "scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153" ! members = [ - "joohank", # (6 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_discord-verify will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_discord-verify" { id = "scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745" ! members = [ - "joohank", # (6 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_groupme-mirror will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_groupme-mirror" { id = "scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39" ! members = [ - "joohank", # (6 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-courses will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-courses" { id = "scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77" ! members = [ - "ckawibow", - "kyma", - "sgli", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-courses_courses will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-courses_courses" { id = "scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406" ! members = [ - "ckawibow", - "kyma", - "sgli", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-housing_housing" { id = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d" ! members = [ - "xindiy", # (13 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-maps_maps will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-maps_maps" { id = "scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64" ! members = [ - "arnavrao", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-study_study will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-study_study" { id = "scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a" ! members = [ - "andyj", - "dsoral", - "klaipras", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops will be updated in-place ! resource "keycloak_group_memberships" "project_devops" { id = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7" ! members = [ - "joohank", - "rjwu", - "saniyav", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_changelog will be updated in-place ! resource "keycloak_group_memberships" "project_devops_changelog" { id = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77" ! members = [ - "joohank", - "rjwu", - "saniyav", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_documentation will be updated in-place ! resource "keycloak_group_memberships" "project_devops_documentation" { id = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f" ! members = [ - "joohank", - "rjwu", - "saniyav", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_link-shortener will be updated in-place ! resource "keycloak_group_memberships" "project_devops_link-shortener" { id = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935" ! members = [ - "joohank", - "rjwu", - "saniyav", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-agent_admins will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-agent_admins" { id = "scottylabs/group-memberships/f4e82f9b-769d-4f38-a7f2-e214189d9269" ! members = [ - "jessec2", - "jiyas", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-mcp-server_admins will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server_admins" { id = "scottylabs/group-memberships/8f84f9ec-e78a-41f6-9ce6-097fce273298" ! members = [ - "jessec2", - "jiyas", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-sms-surface_admins will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface_admins" { id = "scottylabs/group-memberships/34404457-985f-4a61-b2f8-4754c2eb3293" ! members = [ - "jessec2", - "jiyas", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-surface_admins will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-surface_admins" { id = "scottylabs/group-memberships/4f3531db-8075-480e-bbf4-253e98783b56" ! members = [ - "jessec2", - "jiyas", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_tartan-vote will be updated in-place ! resource "keycloak_group_memberships" "project_tartan-vote" { id = "scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c" ! members = [ - "drishtis", - "joohank", - "jzd", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ - "andyj", - "arnavrao", - "ckawibow", - "drishtis", - "dsoral", - "joohank", - "jzd", - "klaipras", - "kyma", - "rjwu", - "saniyav", - "sgli", - "xindiy", # (37 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 19 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 19 to change, 0 to destroy. --- * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
atlantis/plan: google_groups Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: keycloak Plan: 0 to add, 19 to change, 0 to destroy.
atlantis/plan 5/6 projects failed to plan.
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
11aae885db

Ran Plan for 6 projects:

  1. project: google_groups dir: tofu/google_groups workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: git dir: tofu/git workspace: default
  4. project: google_play dir: tofu/google_play workspace: default
  5. project: slack dir: tofu/slack workspace: default
  6. project: keycloak dir: tofu/keycloak workspace: default

1. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #301. To continue, delete the lock from #301 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #301. To continue, delete the lock from #301 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #301. To continue, delete the lock from #301 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #301. To continue, delete the lock from #301 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #301. To continue, delete the lock from #301 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.06s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cbp_bus-sign will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_bus-sign" {
        id       = "scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b"
!       members  = [
-           "joohank",
+           "kalee2",
            # (6 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_dalmatian will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_dalmatian" {
        id       = "scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153"
!       members  = [
-           "joohank",
+           "kalee2",
            # (6 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_discord-verify will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_discord-verify" {
        id       = "scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745"
!       members  = [
-           "joohank",
+           "kalee2",
            # (6 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_groupme-mirror will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_groupme-mirror" {
        id       = "scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39"
!       members  = [
-           "joohank",
+           "kalee2",
            # (6 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-courses will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-courses" {
        id       = "scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77"
!       members  = [
-           "ckawibow",
-           "kyma",
-           "sgli",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-courses_courses will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-courses_courses" {
        id       = "scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406"
!       members  = [
-           "ckawibow",
-           "kyma",
-           "sgli",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-housing_housing" {
        id       = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d"
!       members  = [
-           "xindiy",
            # (13 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-maps_maps will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-maps_maps" {
        id       = "scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64"
!       members  = [
-           "arnavrao",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-study_study will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-study_study" {
        id       = "scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a"
!       members  = [
-           "andyj",
-           "dsoral",
-           "klaipras",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops will be updated in-place
! resource "keycloak_group_memberships" "project_devops" {
        id       = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7"
!       members  = [
-           "joohank",
-           "rjwu",
-           "saniyav",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_changelog will be updated in-place
! resource "keycloak_group_memberships" "project_devops_changelog" {
        id       = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77"
!       members  = [
-           "joohank",
-           "rjwu",
-           "saniyav",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_documentation will be updated in-place
! resource "keycloak_group_memberships" "project_devops_documentation" {
        id       = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f"
!       members  = [
-           "joohank",
-           "rjwu",
-           "saniyav",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_link-shortener will be updated in-place
! resource "keycloak_group_memberships" "project_devops_link-shortener" {
        id       = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935"
!       members  = [
-           "joohank",
-           "rjwu",
-           "saniyav",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-agent_admins will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-agent_admins" {
        id       = "scottylabs/group-memberships/f4e82f9b-769d-4f38-a7f2-e214189d9269"
!       members  = [
-           "jessec2",
-           "jiyas",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-mcp-server_admins will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server_admins" {
        id       = "scottylabs/group-memberships/8f84f9ec-e78a-41f6-9ce6-097fce273298"
!       members  = [
-           "jessec2",
-           "jiyas",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-sms-surface_admins will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface_admins" {
        id       = "scottylabs/group-memberships/34404457-985f-4a61-b2f8-4754c2eb3293"
!       members  = [
-           "jessec2",
-           "jiyas",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-surface_admins will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-surface_admins" {
        id       = "scottylabs/group-memberships/4f3531db-8075-480e-bbf4-253e98783b56"
!       members  = [
-           "jessec2",
-           "jiyas",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_tartan-vote will be updated in-place
! resource "keycloak_group_memberships" "project_tartan-vote" {
        id       = "scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c"
!       members  = [
-           "drishtis",
-           "joohank",
-           "jzd",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
-           "andyj",
-           "arnavrao",
-           "ckawibow",
-           "drishtis",
-           "dsoral",
-           "joohank",
-           "jzd",
-           "klaipras",
-           "kyma",
-           "rjwu",
-           "saniyav",
-           "sgli",
-           "xindiy",
+           "kalee2",
            # (37 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 19 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 19 to change, 0 to destroy.


Plan Summary

6 projects, 1 with changes, 0 with no changes, 5 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #301. To continue, delete the lock from #301 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #301. To continue, delete the lock from #301 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #301. To continue, delete the lock from #301 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #301. To continue, delete the lock from #301 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #301. To continue, delete the lock from #301 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.06s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cbp_bus-sign will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_bus-sign" { id = "scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b" ! members = [ - "joohank", + "kalee2", # (6 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_dalmatian will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_dalmatian" { id = "scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153" ! members = [ - "joohank", + "kalee2", # (6 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_discord-verify will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_discord-verify" { id = "scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745" ! members = [ - "joohank", + "kalee2", # (6 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_groupme-mirror will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_groupme-mirror" { id = "scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39" ! members = [ - "joohank", + "kalee2", # (6 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-courses will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-courses" { id = "scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77" ! members = [ - "ckawibow", - "kyma", - "sgli", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-courses_courses will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-courses_courses" { id = "scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406" ! members = [ - "ckawibow", - "kyma", - "sgli", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-housing_housing" { id = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d" ! members = [ - "xindiy", # (13 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-maps_maps will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-maps_maps" { id = "scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64" ! members = [ - "arnavrao", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-study_study will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-study_study" { id = "scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a" ! members = [ - "andyj", - "dsoral", - "klaipras", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops will be updated in-place ! resource "keycloak_group_memberships" "project_devops" { id = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7" ! members = [ - "joohank", - "rjwu", - "saniyav", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_changelog will be updated in-place ! resource "keycloak_group_memberships" "project_devops_changelog" { id = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77" ! members = [ - "joohank", - "rjwu", - "saniyav", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_documentation will be updated in-place ! resource "keycloak_group_memberships" "project_devops_documentation" { id = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f" ! members = [ - "joohank", - "rjwu", - "saniyav", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_link-shortener will be updated in-place ! resource "keycloak_group_memberships" "project_devops_link-shortener" { id = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935" ! members = [ - "joohank", - "rjwu", - "saniyav", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-agent_admins will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-agent_admins" { id = "scottylabs/group-memberships/f4e82f9b-769d-4f38-a7f2-e214189d9269" ! members = [ - "jessec2", - "jiyas", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-mcp-server_admins will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server_admins" { id = "scottylabs/group-memberships/8f84f9ec-e78a-41f6-9ce6-097fce273298" ! members = [ - "jessec2", - "jiyas", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-sms-surface_admins will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface_admins" { id = "scottylabs/group-memberships/34404457-985f-4a61-b2f8-4754c2eb3293" ! members = [ - "jessec2", - "jiyas", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-surface_admins will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-surface_admins" { id = "scottylabs/group-memberships/4f3531db-8075-480e-bbf4-253e98783b56" ! members = [ - "jessec2", - "jiyas", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_tartan-vote will be updated in-place ! resource "keycloak_group_memberships" "project_tartan-vote" { id = "scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c" ! members = [ - "drishtis", - "joohank", - "jzd", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ - "andyj", - "arnavrao", - "ckawibow", - "drishtis", - "dsoral", - "joohank", - "jzd", - "klaipras", - "kyma", - "rjwu", - "saniyav", - "sgli", - "xindiy", + "kalee2", # (37 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 19 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 19 to change, 0 to destroy. --- ### Plan Summary 6 projects, 1 with changes, 0 with no changes, 5 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Author
Member

Proof of linkage

Proof of linkage
Member

atlantis unlock

atlantis unlock

All Atlantis locks for this PR have been unlocked and plans discarded

All Atlantis locks for this PR have been unlocked and plans discarded
tentype approved these changes 2026-09-20 18:48:38 +00:00
xboxbedrock force-pushed main from 11aae885db
Some checks failed
atlantis/plan: google_groups Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: keycloak Plan: 0 to add, 19 to change, 0 to destroy.
atlantis/plan 5/6 projects failed to plan.
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
to c0b3387df1
All checks were successful
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: slack Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 5 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
2026-09-21 03:52:41 +00:00
Compare

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.64s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_cbp_ch0_kalee2 will be created
+ resource "null_resource" "slack_cbp_ch0_kalee2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08S56CJ8R2"
          + "user"    = "U0C15673GG5"
        }
    }

  # null_resource.slack_hub_kalee2 will be created
+ resource "null_resource" "slack_hub_kalee2" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0C15673GG5"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 0 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.60s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.60s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_kalee2 will be created
+ resource "google_cloud_identity_group_membership" "tech_kalee2" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "kalee2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.64s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cbp_bus-sign will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_bus-sign" {
        id       = "scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b"
!       members  = [
+           "kalee2",
            # (14 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_dalmatian will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_dalmatian" {
        id       = "scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153"
!       members  = [
+           "kalee2",
            # (14 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_discord-verify will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_discord-verify" {
        id       = "scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745"
!       members  = [
+           "kalee2",
            # (14 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_groupme-mirror will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_groupme-mirror" {
        id       = "scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39"
!       members  = [
+           "kalee2",
            # (14 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "kalee2",
            # (74 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 5 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 5 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.59s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.kalee2 will be created
+ resource "discord_member_roles" "kalee2" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "688872544560087135"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1494773174334783499"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/267/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/267/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/267/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 7.61s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # forgejo_team_member.cbp_kalee2 will be created
+ resource "forgejo_team_member" "cbp_kalee2" {
+       team_id = 3
+       user    = "kalee2"
    }

  # github_team_membership.cbp_kalee2 will be created
+ resource "github_team_membership" "cbp_kalee2" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "17170099"
+       username = "kalee88"
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 0 to change, 0 to destroy.


Plan Summary

6 projects, 5 with changes, 1 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.64s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_cbp_ch0_kalee2 will be created + resource "null_resource" "slack_cbp_ch0_kalee2" { + id = (known after apply) + triggers = { + "channel" = "C08S56CJ8R2" + "user" = "U0C15673GG5" } } # null_resource.slack_hub_kalee2 will be created + resource "null_resource" "slack_hub_kalee2" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0C15673GG5" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.60s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.60s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_kalee2 will be created + resource "google_cloud_identity_group_membership" "tech_kalee2" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "kalee2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.64s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cbp_bus-sign will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_bus-sign" { id = "scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b" ! members = [ + "kalee2", # (14 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_dalmatian will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_dalmatian" { id = "scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153" ! members = [ + "kalee2", # (14 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_discord-verify will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_discord-verify" { id = "scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745" ! members = [ + "kalee2", # (14 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_groupme-mirror will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_groupme-mirror" { id = "scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39" ! members = [ + "kalee2", # (14 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "kalee2", # (74 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 5 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 5 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.59s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.kalee2 will be created + resource "discord_member_roles" "kalee2" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "688872544560087135" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1494773174334783499" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/267/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/267/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/267/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 7.61s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # forgejo_team_member.cbp_kalee2 will be created + resource "forgejo_team_member" "cbp_kalee2" { + team_id = 3 + user = "kalee2" } # github_team_membership.cbp_kalee2 will be created + resource "github_team_membership" "cbp_kalee2" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "17170099" + username = "kalee88" } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### Plan Summary 6 projects, 5 with changes, 1 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis apply

atlantis apply

Ran Apply for 6 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: keycloak dir: tofu/keycloak workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.25s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

2. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.15s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_cbp_ch0_kalee2: Creating...
null_resource.slack_hub_kalee2: Creating...
null_resource.slack_cbp_ch0_kalee2: Provisioning with 'local-exec'...
null_resource.slack_cbp_ch0_kalee2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08S56CJ8R2 --user U0C15673GG5"]
null_resource.slack_hub_kalee2: Provisioning with 'local-exec'...
null_resource.slack_hub_kalee2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C15673GG5"]
null_resource.slack_hub_kalee2: Creation complete after 0s [id=7357239019431343470]
null_resource.slack_cbp_ch0_kalee2: Creation complete after 0s [id=4795150684548876640]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

3. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.kalee2: Creating...
discord_member_roles.kalee2: Creation complete after 1s [id=1353118209024331948:688872544560087135]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.17s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_kalee2: Creating...
google_cloud_identity_group_membership.tech_kalee2: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/116836211437733237233]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.22s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
github_team_membership.cbp_kalee2: Creating...
forgejo_team_member.cbp_kalee2: Creating...
forgejo_team_member.cbp_kalee2: Creation complete after 0s
github_team_membership.cbp_kalee2: Creation complete after 1s [id=17170099:kalee88]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

6. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.30s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cbp_discord-verify: Modifying... [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745]
keycloak_group_memberships.project_cbp_groupme-mirror: Modifying... [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39]
keycloak_group_memberships.project_cbp_bus-sign: Modifying... [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b]
keycloak_group_memberships.project_cbp_dalmatian: Modifying... [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_cbp_discord-verify: Modifications complete after 2s [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745]
keycloak_group_memberships.project_cbp_dalmatian: Modifications complete after 2s [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153]
keycloak_group_memberships.project_cbp_groupme-mirror: Modifications complete after 2s [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39]
keycloak_group_memberships.project_cbp_bus-sign: Modifications complete after 2s [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b]
keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 5 changed, 0 destroyed.

Apply Summary

6 projects, 6 successful, 0 failed, 0 errored

Ran Apply for 6 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.25s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.15s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_cbp_ch0_kalee2: Creating... null_resource.slack_hub_kalee2: Creating... null_resource.slack_cbp_ch0_kalee2: Provisioning with 'local-exec'... null_resource.slack_cbp_ch0_kalee2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08S56CJ8R2 --user U0C15673GG5"] null_resource.slack_hub_kalee2: Provisioning with 'local-exec'... null_resource.slack_hub_kalee2 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C15673GG5"] null_resource.slack_hub_kalee2: Creation complete after 0s [id=7357239019431343470] null_resource.slack_cbp_ch0_kalee2: Creation complete after 0s [id=4795150684548876640] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.kalee2: Creating... discord_member_roles.kalee2: Creation complete after 1s [id=1353118209024331948:688872544560087135] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.17s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_kalee2: Creating... google_cloud_identity_group_membership.tech_kalee2: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/116836211437733237233] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.22s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ github_team_membership.cbp_kalee2: Creating... forgejo_team_member.cbp_kalee2: Creating... forgejo_team_member.cbp_kalee2: Creation complete after 0s github_team_membership.cbp_kalee2: Creation complete after 1s [id=17170099:kalee88] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 6. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.30s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cbp_discord-verify: Modifying... [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745] keycloak_group_memberships.project_cbp_groupme-mirror: Modifying... [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39] keycloak_group_memberships.project_cbp_bus-sign: Modifying... [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b] keycloak_group_memberships.project_cbp_dalmatian: Modifying... [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_cbp_discord-verify: Modifications complete after 2s [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745] keycloak_group_memberships.project_cbp_dalmatian: Modifications complete after 2s [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153] keycloak_group_memberships.project_cbp_groupme-mirror: Modifications complete after 2s [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39] keycloak_group_memberships.project_cbp_bus-sign: Modifications complete after 2s [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b] keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 5 changed, 0 destroyed. ``` </details> --- ### Apply Summary 6 projects, 6 successful, 0 failed, 0 errored
xboxbedrock referenced this pull request from a commit 2026-09-21 03:55:12 +00:00

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
brightz referenced this pull request from a commit 2026-09-26 22:03:28 +00:00
All checks were successful
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: slack Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 5 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
Some workflows are waiting to be reviewed.
Sign in to join this conversation.
No description provided.