feat: add xindi to housing #250

Merged
xboxbedrock merged 2 commits from xindiy/governance:main into main 2026-09-19 20:59:29 +00:00
Member
No description provided.
Update data/teams/cmu-housing.toml
All checks were successful
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
CI / build (pull_request) Has been skipped
Validate / Governance (pull_request_target) Successful in 8m54s
CI / check-1 (pull_request) Successful in 13m12s
CI / check (pull_request) Successful in 0s
93cfa40a34
NikhilR approved these changes 2026-09-19 20:34:25 +00:00
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 1 to add, 0 to change, 6 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 2 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 6 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 2 to destroy.
atlantis/plan: git Plan: 2 to add, 0 to change, 4 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
4fcd70fb4b

Ran Plan for 6 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: git dir: tofu/git workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: discord dir: tofu/discord workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default
  6. project: slack dir: tofu/slack workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.26s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_devops_ch0_aszhou will be destroyed
  # (because null_resource.slack_devops_ch0_aszhou is not in configuration)
- resource "null_resource" "slack_devops_ch0_aszhou" {
-       id       = "528167437641880870" -> null
-       triggers = {
          - "channel" = "C08K3Q77ZQF"
          - "user"    = "U09CQ04V4BT"
        } -> null
    }

  # null_resource.slack_devops_ch0_indigo0445 will be destroyed
  # (because null_resource.slack_devops_ch0_indigo0445 is not in configuration)
- resource "null_resource" "slack_devops_ch0_indigo0445" {
-       id       = "6369468938332311989" -> null
-       triggers = {
          - "channel" = "C08K3Q77ZQF"
          - "user"    = "U09C4RTR36F"
        } -> null
    }

  # null_resource.slack_devops_ch1_aszhou will be destroyed
  # (because null_resource.slack_devops_ch1_aszhou is not in configuration)
- resource "null_resource" "slack_devops_ch1_aszhou" {
-       id       = "2598782963452480140" -> null
-       triggers = {
          - "channel" = "C0BNV84NUTC"
          - "user"    = "U09CQ04V4BT"
        } -> null
    }

  # null_resource.slack_devops_ch1_indigo0445 will be destroyed
  # (because null_resource.slack_devops_ch1_indigo0445 is not in configuration)
- resource "null_resource" "slack_devops_ch1_indigo0445" {
-       id       = "4030560679133036267" -> null
-       triggers = {
          - "channel" = "C0BNV84NUTC"
          - "user"    = "U09C4RTR36F"
        } -> null
    }

  # null_resource.slack_hub_aszhou will be destroyed
  # (because null_resource.slack_hub_aszhou is not in configuration)
- resource "null_resource" "slack_hub_aszhou" {
-       id       = "7120216021093146134" -> null
-       triggers = {
          - "channel" = "C096TM8EMS8"
          - "user"    = "U09CQ04V4BT"
        } -> null
    }

  # null_resource.slack_hub_indigo0445 will be destroyed
  # (because null_resource.slack_hub_indigo0445 is not in configuration)
- resource "null_resource" "slack_hub_indigo0445" {
-       id       = "7973140652129308373" -> null
-       triggers = {
          - "channel" = "C096TM8EMS8"
          - "user"    = "U09C4RTR36F"
        } -> null
    }

  # null_resource.slack_hub_xindiy will be created
+ resource "null_resource" "slack_hub_xindiy" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0C1AAR21L5"
        }
    }

Plan: 1 to add, 0 to change, 6 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 1 to add, 0 to change, 6 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.22s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_aszhou will be destroyed
  # (because google_cloud_identity_group_membership.tech_aszhou is not in configuration)
- resource "google_cloud_identity_group_membership" "tech_aszhou" {
-       create_time = "2026-09-19T20:39:12.556017Z" -> null
-       group       = "groups/04anzqyu1bqbv79" -> null
-       id          = "groups/04anzqyu1bqbv79/memberships/100575886719005530375" -> null
-       name        = "groups/04anzqyu1bqbv79/memberships/100575886719005530375" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-19T20:39:12.556017Z" -> null

-       preferred_member_key {
-           id = "aszhou@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.tech_indigo0445 will be destroyed
  # (because google_cloud_identity_group_membership.tech_indigo0445 is not in configuration)
- resource "google_cloud_identity_group_membership" "tech_indigo0445" {
-       create_time = "2026-09-19T20:50:47.649232Z" -> null
-       group       = "groups/04anzqyu1bqbv79" -> null
-       id          = "groups/04anzqyu1bqbv79/memberships/100660819832913404095" -> null
-       name        = "groups/04anzqyu1bqbv79/memberships/100660819832913404095" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-19T20:50:47.649232Z" -> null

-       preferred_member_key {
-           id = "rjwu@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.tech_xindiy will be created
+ resource "google_cloud_identity_group_membership" "tech_xindiy" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "xindiy@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 2 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.27s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-housing_housing" {
        id       = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d"
!       members  = [
+           "xindiy",
            # (13 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops will be updated in-place
! resource "keycloak_group_memberships" "project_devops" {
        id       = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7"
!       members  = [
-           "aszhou",
-           "rjwu",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_changelog will be updated in-place
! resource "keycloak_group_memberships" "project_devops_changelog" {
        id       = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77"
!       members  = [
-           "aszhou",
-           "rjwu",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_documentation will be updated in-place
! resource "keycloak_group_memberships" "project_devops_documentation" {
        id       = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f"
!       members  = [
-           "aszhou",
-           "rjwu",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_link-shortener will be updated in-place
! resource "keycloak_group_memberships" "project_devops_link-shortener" {
        id       = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935"
!       members  = [
-           "aszhou",
-           "rjwu",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
-           "aszhou",
-           "rjwu",
+           "xindiy",
            # (36 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 6 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 6 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.27s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # discord_member_roles.aszhou will be destroyed
  # (because discord_member_roles.aszhou is not in configuration)
- resource "discord_member_roles" "aszhou" {
-       id        = "1353118209024331948:572567442153406479" -> null
-       server_id = "1353118209024331948" -> null
-       user_id   = "572567442153406479" -> null

-       role {
-           has_role = true -> null
-           role_id  = "1488787087598813215" -> null
        }
-       role {
-           has_role = true -> null
-           role_id  = "1488787088118644848" -> null
        }
    }

  # discord_member_roles.indigo0445 will be destroyed
  # (because discord_member_roles.indigo0445 is not in configuration)
- resource "discord_member_roles" "indigo0445" {
-       id        = "1353118209024331948:629857326232371210" -> null
-       server_id = "1353118209024331948" -> null
-       user_id   = "629857326232371210" -> null

-       role {
-           has_role = true -> null
-           role_id  = "1488787087598813215" -> null
        }
-       role {
-           has_role = true -> null
-           role_id  = "1488787088118644848" -> null
        }
    }

  # discord_member_roles.xindiy will be created
+ resource "discord_member_roles" "xindiy" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "712363878851280936"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1513653499957874871"
        }
    }

Plan: 1 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 2 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on shared package cache
    Finished `release` profile [optimized] target(s) in 0.22s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # forgejo_team_member.cmu-housing_xindiy will be created
+ resource "forgejo_team_member" "cmu-housing_xindiy" {
+       team_id = 2
+       user    = "xindiy"
    }

  # forgejo_team_member.devops_aszhou will be destroyed
  # (because forgejo_team_member.devops_aszhou is not in configuration)
- resource "forgejo_team_member" "devops_aszhou" {
-       team_id = 4 -> null
-       user    = "aszhou" -> null
    }

  # forgejo_team_member.devops_indigo0445 will be destroyed
  # (because forgejo_team_member.devops_indigo0445 is not in configuration)
- resource "forgejo_team_member" "devops_indigo0445" {
-       team_id = 4 -> null
-       user    = "rjwu" -> null
    }

  # github_team_membership.cmu-housing_xindiy will be created
+ resource "github_team_membership" "cmu-housing_xindiy" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "17937919"
+       username = "xindiyang0"
    }

  # github_team_membership.devops_aszhou will be destroyed
  # (because github_team_membership.devops_aszhou is not in configuration)
- resource "github_team_membership" "devops_aszhou" {
-       etag     = "W/\"7ded4c758e7d683f26bbd6ba5249b0169eb642c18bd932c07b551232053c34d8\"" -> null
-       id       = "14706218:azhou22" -> null
-       role     = "member" -> null
-       team_id  = "14706218" -> null
-       username = "azhou22" -> null
    }

  # github_team_membership.devops_indigo0445 will be destroyed
  # (because github_team_membership.devops_indigo0445 is not in configuration)
- resource "github_team_membership" "devops_indigo0445" {
-       etag     = "W/\"b583553f0de8ce760b6183f9442faaf221b0085c93ef475142722a4d1f324c8b\"" -> null
-       id       = "14706218:indigo0445" -> null
-       role     = "member" -> null
-       team_id  = "14706218" -> null
-       username = "indigo0445" -> null
    }

Plan: 2 to add, 0 to change, 4 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 0 to change, 4 to destroy.


Plan Summary

6 projects, 5 with changes, 1 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.26s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # null_resource.slack_devops_ch0_aszhou will be destroyed # (because null_resource.slack_devops_ch0_aszhou is not in configuration) - resource "null_resource" "slack_devops_ch0_aszhou" { - id = "528167437641880870" -> null - triggers = { - "channel" = "C08K3Q77ZQF" - "user" = "U09CQ04V4BT" } -> null } # null_resource.slack_devops_ch0_indigo0445 will be destroyed # (because null_resource.slack_devops_ch0_indigo0445 is not in configuration) - resource "null_resource" "slack_devops_ch0_indigo0445" { - id = "6369468938332311989" -> null - triggers = { - "channel" = "C08K3Q77ZQF" - "user" = "U09C4RTR36F" } -> null } # null_resource.slack_devops_ch1_aszhou will be destroyed # (because null_resource.slack_devops_ch1_aszhou is not in configuration) - resource "null_resource" "slack_devops_ch1_aszhou" { - id = "2598782963452480140" -> null - triggers = { - "channel" = "C0BNV84NUTC" - "user" = "U09CQ04V4BT" } -> null } # null_resource.slack_devops_ch1_indigo0445 will be destroyed # (because null_resource.slack_devops_ch1_indigo0445 is not in configuration) - resource "null_resource" "slack_devops_ch1_indigo0445" { - id = "4030560679133036267" -> null - triggers = { - "channel" = "C0BNV84NUTC" - "user" = "U09C4RTR36F" } -> null } # null_resource.slack_hub_aszhou will be destroyed # (because null_resource.slack_hub_aszhou is not in configuration) - resource "null_resource" "slack_hub_aszhou" { - id = "7120216021093146134" -> null - triggers = { - "channel" = "C096TM8EMS8" - "user" = "U09CQ04V4BT" } -> null } # null_resource.slack_hub_indigo0445 will be destroyed # (because null_resource.slack_hub_indigo0445 is not in configuration) - resource "null_resource" "slack_hub_indigo0445" { - id = "7973140652129308373" -> null - triggers = { - "channel" = "C096TM8EMS8" - "user" = "U09C4RTR36F" } -> null } # null_resource.slack_hub_xindiy will be created + resource "null_resource" "slack_hub_xindiy" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0C1AAR21L5" } } Plan: 1 to add, 0 to change, 6 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 1 to add, 0 to change, 6 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.22s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_aszhou will be destroyed # (because google_cloud_identity_group_membership.tech_aszhou is not in configuration) - resource "google_cloud_identity_group_membership" "tech_aszhou" { - create_time = "2026-09-19T20:39:12.556017Z" -> null - group = "groups/04anzqyu1bqbv79" -> null - id = "groups/04anzqyu1bqbv79/memberships/100575886719005530375" -> null - name = "groups/04anzqyu1bqbv79/memberships/100575886719005530375" -> null - type = "USER" -> null - update_time = "2026-09-19T20:39:12.556017Z" -> null - preferred_member_key { - id = "aszhou@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.tech_indigo0445 will be destroyed # (because google_cloud_identity_group_membership.tech_indigo0445 is not in configuration) - resource "google_cloud_identity_group_membership" "tech_indigo0445" { - create_time = "2026-09-19T20:50:47.649232Z" -> null - group = "groups/04anzqyu1bqbv79" -> null - id = "groups/04anzqyu1bqbv79/memberships/100660819832913404095" -> null - name = "groups/04anzqyu1bqbv79/memberships/100660819832913404095" -> null - type = "USER" -> null - update_time = "2026-09-19T20:50:47.649232Z" -> null - preferred_member_key { - id = "rjwu@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.tech_xindiy will be created + resource "google_cloud_identity_group_membership" "tech_xindiy" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "xindiy@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 2 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.27s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-housing_housing" { id = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d" ! members = [ + "xindiy", # (13 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops will be updated in-place ! resource "keycloak_group_memberships" "project_devops" { id = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7" ! members = [ - "aszhou", - "rjwu", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_changelog will be updated in-place ! resource "keycloak_group_memberships" "project_devops_changelog" { id = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77" ! members = [ - "aszhou", - "rjwu", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_documentation will be updated in-place ! resource "keycloak_group_memberships" "project_devops_documentation" { id = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f" ! members = [ - "aszhou", - "rjwu", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_link-shortener will be updated in-place ! resource "keycloak_group_memberships" "project_devops_link-shortener" { id = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935" ! members = [ - "aszhou", - "rjwu", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ - "aszhou", - "rjwu", + "xindiy", # (36 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 6 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 6 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.27s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # discord_member_roles.aszhou will be destroyed # (because discord_member_roles.aszhou is not in configuration) - resource "discord_member_roles" "aszhou" { - id = "1353118209024331948:572567442153406479" -> null - server_id = "1353118209024331948" -> null - user_id = "572567442153406479" -> null - role { - has_role = true -> null - role_id = "1488787087598813215" -> null } - role { - has_role = true -> null - role_id = "1488787088118644848" -> null } } # discord_member_roles.indigo0445 will be destroyed # (because discord_member_roles.indigo0445 is not in configuration) - resource "discord_member_roles" "indigo0445" { - id = "1353118209024331948:629857326232371210" -> null - server_id = "1353118209024331948" -> null - user_id = "629857326232371210" -> null - role { - has_role = true -> null - role_id = "1488787087598813215" -> null } - role { - has_role = true -> null - role_id = "1488787088118644848" -> null } } # discord_member_roles.xindiy will be created + resource "discord_member_roles" "xindiy" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "712363878851280936" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1513653499957874871" } } Plan: 1 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 2 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on shared package cache Finished `release` profile [optimized] target(s) in 0.22s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # forgejo_team_member.cmu-housing_xindiy will be created + resource "forgejo_team_member" "cmu-housing_xindiy" { + team_id = 2 + user = "xindiy" } # forgejo_team_member.devops_aszhou will be destroyed # (because forgejo_team_member.devops_aszhou is not in configuration) - resource "forgejo_team_member" "devops_aszhou" { - team_id = 4 -> null - user = "aszhou" -> null } # forgejo_team_member.devops_indigo0445 will be destroyed # (because forgejo_team_member.devops_indigo0445 is not in configuration) - resource "forgejo_team_member" "devops_indigo0445" { - team_id = 4 -> null - user = "rjwu" -> null } # github_team_membership.cmu-housing_xindiy will be created + resource "github_team_membership" "cmu-housing_xindiy" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "17937919" + username = "xindiyang0" } # github_team_membership.devops_aszhou will be destroyed # (because github_team_membership.devops_aszhou is not in configuration) - resource "github_team_membership" "devops_aszhou" { - etag = "W/\"7ded4c758e7d683f26bbd6ba5249b0169eb642c18bd932c07b551232053c34d8\"" -> null - id = "14706218:azhou22" -> null - role = "member" -> null - team_id = "14706218" -> null - username = "azhou22" -> null } # github_team_membership.devops_indigo0445 will be destroyed # (because github_team_membership.devops_indigo0445 is not in configuration) - resource "github_team_membership" "devops_indigo0445" { - etag = "W/\"b583553f0de8ce760b6183f9442faaf221b0085c93ef475142722a4d1f324c8b\"" -> null - id = "14706218:indigo0445" -> null - role = "member" -> null - team_id = "14706218" -> null - username = "indigo0445" -> null } Plan: 2 to add, 0 to change, 4 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 0 to change, 4 to destroy. --- ### Plan Summary 6 projects, 5 with changes, 1 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis apply

atlantis apply

Ran Apply for 6 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: keycloak dir: tofu/keycloak workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.20s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

2. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.27s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_hub_indigo0445: Destroying... [id=7973140652129308373]
null_resource.slack_devops_ch1_aszhou: Destroying... [id=2598782963452480140]
null_resource.slack_devops_ch1_aszhou: Destruction complete after 0s
null_resource.slack_hub_aszhou: Destroying... [id=7120216021093146134]
null_resource.slack_devops_ch0_indigo0445: Destroying... [id=6369468938332311989]
null_resource.slack_devops_ch0_indigo0445: Destruction complete after 0s
null_resource.slack_hub_aszhou: Destruction complete after 0s
null_resource.slack_devops_ch1_indigo0445: Destroying... [id=4030560679133036267]
null_resource.slack_devops_ch0_aszhou: Destroying... [id=528167437641880870]
null_resource.slack_hub_indigo0445: Destruction complete after 0s
null_resource.slack_devops_ch1_indigo0445: Destruction complete after 0s
null_resource.slack_devops_ch0_aszhou: Destruction complete after 0s
null_resource.slack_hub_xindiy: Creating...
null_resource.slack_hub_xindiy: Provisioning with 'local-exec'...
null_resource.slack_hub_xindiy (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C1AAR21L5"]
null_resource.slack_hub_xindiy: Creation complete after 0s [id=1011018451728934545]

Apply complete! Resources: 1 added, 0 changed, 6 destroyed.

3. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.21s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.indigo0445: Destroying... [id=1353118209024331948:629857326232371210]
discord_member_roles.aszhou: Destroying... [id=1353118209024331948:572567442153406479]
discord_member_roles.xindiy: Creating...
discord_member_roles.indigo0445: Destruction complete after 0s
discord_member_roles.aszhou: Destruction complete after 1s
discord_member_roles.xindiy: Creation complete after 1s [id=1353118209024331948:712363878851280936]

Apply complete! Resources: 1 added, 0 changed, 2 destroyed.

4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.20s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_indigo0445: Destroying... [id=groups/04anzqyu1bqbv79/memberships/100660819832913404095]
google_cloud_identity_group_membership.tech_aszhou: Destroying... [id=groups/04anzqyu1bqbv79/memberships/100575886719005530375]
google_cloud_identity_group_membership.tech_xindiy: Creating...
google_cloud_identity_group_membership.tech_aszhou: Destruction complete after 0s
google_cloud_identity_group_membership.tech_indigo0445: Destruction complete after 1s
google_cloud_identity_group_membership.tech_xindiy: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/100687050859054942101]

Apply complete! Resources: 1 added, 0 changed, 2 destroyed.

5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.24s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
github_team_membership.devops_aszhou: Destroying... [id=14706218:azhou22]
github_team_membership.devops_indigo0445: Destroying... [id=14706218:indigo0445]
github_team_membership.cmu-housing_xindiy: Creating...
forgejo_team_member.devops_indigo0445: Destroying...
forgejo_team_member.devops_aszhou: Destroying...
forgejo_team_member.cmu-housing_xindiy: Creating...
forgejo_team_member.cmu-housing_xindiy: Creation complete after 1s
forgejo_team_member.devops_aszhou: Destruction complete after 1s
forgejo_team_member.devops_indigo0445: Destruction complete after 1s
github_team_membership.devops_indigo0445: Destruction complete after 1s
github_team_membership.devops_aszhou: Destruction complete after 2s
github_team_membership.cmu-housing_xindiy: Creation complete after 4s [id=17937919:xindiyang0]

Apply complete! Resources: 2 added, 0 changed, 4 destroyed.

6. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.26s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_devops_link-shortener: Modifying... [id=scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935]
keycloak_group_memberships.project_devops_changelog: Modifying... [id=scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77]
keycloak_group_memberships.project_devops_documentation: Modifying... [id=scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f]
keycloak_group_memberships.project_cmu-housing_housing: Modifying... [id=scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d]
keycloak_group_memberships.project_devops: Modifying... [id=scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_devops_link-shortener: Modifications complete after 0s [id=scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935]
keycloak_group_memberships.project_devops_documentation: Modifications complete after 0s [id=scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f]
keycloak_group_memberships.project_devops_changelog: Modifications complete after 0s [id=scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77]
keycloak_group_memberships.project_devops: Modifications complete after 0s [id=scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7]
keycloak_group_memberships.project_cmu-housing_housing: Modifications complete after 2s [id=scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d]
keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 6 changed, 0 destroyed.

Apply Summary

6 projects, 6 successful, 0 failed, 0 errored

Ran Apply for 6 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.20s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.27s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_hub_indigo0445: Destroying... [id=7973140652129308373] null_resource.slack_devops_ch1_aszhou: Destroying... [id=2598782963452480140] null_resource.slack_devops_ch1_aszhou: Destruction complete after 0s null_resource.slack_hub_aszhou: Destroying... [id=7120216021093146134] null_resource.slack_devops_ch0_indigo0445: Destroying... [id=6369468938332311989] null_resource.slack_devops_ch0_indigo0445: Destruction complete after 0s null_resource.slack_hub_aszhou: Destruction complete after 0s null_resource.slack_devops_ch1_indigo0445: Destroying... [id=4030560679133036267] null_resource.slack_devops_ch0_aszhou: Destroying... [id=528167437641880870] null_resource.slack_hub_indigo0445: Destruction complete after 0s null_resource.slack_devops_ch1_indigo0445: Destruction complete after 0s null_resource.slack_devops_ch0_aszhou: Destruction complete after 0s null_resource.slack_hub_xindiy: Creating... null_resource.slack_hub_xindiy: Provisioning with 'local-exec'... null_resource.slack_hub_xindiy (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C1AAR21L5"] null_resource.slack_hub_xindiy: Creation complete after 0s [id=1011018451728934545] Apply complete! Resources: 1 added, 0 changed, 6 destroyed. ``` </details> --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.21s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.indigo0445: Destroying... [id=1353118209024331948:629857326232371210] discord_member_roles.aszhou: Destroying... [id=1353118209024331948:572567442153406479] discord_member_roles.xindiy: Creating... discord_member_roles.indigo0445: Destruction complete after 0s discord_member_roles.aszhou: Destruction complete after 1s discord_member_roles.xindiy: Creation complete after 1s [id=1353118209024331948:712363878851280936] Apply complete! Resources: 1 added, 0 changed, 2 destroyed. ``` </details> --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.20s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_indigo0445: Destroying... [id=groups/04anzqyu1bqbv79/memberships/100660819832913404095] google_cloud_identity_group_membership.tech_aszhou: Destroying... [id=groups/04anzqyu1bqbv79/memberships/100575886719005530375] google_cloud_identity_group_membership.tech_xindiy: Creating... google_cloud_identity_group_membership.tech_aszhou: Destruction complete after 0s google_cloud_identity_group_membership.tech_indigo0445: Destruction complete after 1s google_cloud_identity_group_membership.tech_xindiy: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/100687050859054942101] Apply complete! Resources: 1 added, 0 changed, 2 destroyed. ``` </details> --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.24s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ github_team_membership.devops_aszhou: Destroying... [id=14706218:azhou22] github_team_membership.devops_indigo0445: Destroying... [id=14706218:indigo0445] github_team_membership.cmu-housing_xindiy: Creating... forgejo_team_member.devops_indigo0445: Destroying... forgejo_team_member.devops_aszhou: Destroying... forgejo_team_member.cmu-housing_xindiy: Creating... forgejo_team_member.cmu-housing_xindiy: Creation complete after 1s forgejo_team_member.devops_aszhou: Destruction complete after 1s forgejo_team_member.devops_indigo0445: Destruction complete after 1s github_team_membership.devops_indigo0445: Destruction complete after 1s github_team_membership.devops_aszhou: Destruction complete after 2s github_team_membership.cmu-housing_xindiy: Creation complete after 4s [id=17937919:xindiyang0] Apply complete! Resources: 2 added, 0 changed, 4 destroyed. ``` </details> --- ### 6. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.26s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_devops_link-shortener: Modifying... [id=scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935] keycloak_group_memberships.project_devops_changelog: Modifying... [id=scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77] keycloak_group_memberships.project_devops_documentation: Modifying... [id=scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f] keycloak_group_memberships.project_cmu-housing_housing: Modifying... [id=scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d] keycloak_group_memberships.project_devops: Modifying... [id=scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_devops_link-shortener: Modifications complete after 0s [id=scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935] keycloak_group_memberships.project_devops_documentation: Modifications complete after 0s [id=scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f] keycloak_group_memberships.project_devops_changelog: Modifications complete after 0s [id=scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77] keycloak_group_memberships.project_devops: Modifications complete after 0s [id=scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7] keycloak_group_memberships.project_cmu-housing_housing: Modifications complete after 2s [id=scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d] keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 6 changed, 0 destroyed. ``` </details> --- ### Apply Summary 6 projects, 6 successful, 0 failed, 0 errored
xboxbedrock referenced this pull request from a commit 2026-09-19 20:59:29 +00:00

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
brightz referenced this pull request from a commit 2026-09-26 22:03:28 +00:00
Sign in to join this conversation.
No description provided.