feat: add jessica-lr-song to slai #270
No reviewers
Labels
No labels
needs/linking
needs/docs
needs/research
needs/upstream
type/bug
type/external
type/feature
type/refactor
urgency
high
urgency
immediate
urgency
low
urgency
medium
No milestone
No project
No assignees
4 participants
Notifications
Due date
No due date set.
Dependencies
No dependencies set
Reference
ScottyLabs/governance!270
Loading…
Reference in a new issue
No description provided.
Delete branch "jessica-lr-song/governance:main"
Deleting a branch is permanent. Although the deleted branch may continue to exist for a short time before it actually gets removed, it CANNOT be undone in most cases. Continue?
Wait I have no idea what the other stuff is but i just wanted to add myself to slai
Ran Plan for 6 projects:
slackdir:tofu/slackworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:defaultgoogle_playdir:tofu/google_playworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.6. project:
google_playdir:tofu/google_playworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #249. To continue, delete the lock from #249 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.Plan Summary
6 projects, 0 with changes, 0 with no changes, 6 failed
Add Jessica to slaito feat: add jessica to slaiRan Plan for 6 projects:
google_groupsdir:tofu/google_groupsworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgoogle_playdir:tofu/google_playworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultslackdir:tofu/slackworkspace:default1. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #250. To continue, delete the lock from #250 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #250. To continue, delete the lock from #250 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #250. To continue, delete the lock from #250 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
google_playdir:tofu/google_playworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #250. To continue, delete the lock from #250 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #250. To continue, delete the lock from #250 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.6. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #250. To continue, delete the lock from #250 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.Plan Summary
6 projects, 0 with changes, 0 with no changes, 6 failed
feat: add jessica to slaito feat: add jessica-lr-song to slaiRan Plan for 6 projects:
keycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:defaultslackdir:tofu/slackworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultgoogle_playdir:tofu/google_playworkspace:default1. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #250. To continue, delete the lock from #250 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #250. To continue, delete the lock from #250 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #250. To continue, delete the lock from #250 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #250. To continue, delete the lock from #250 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #250. To continue, delete the lock from #250 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.6. project:
google_playdir:tofu/google_playworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #250. To continue, delete the lock from #250 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.Plan Summary
6 projects, 0 with changes, 0 with no changes, 6 failed
Ran Plan for 6 projects:
discorddir:tofu/discordworkspace:defaultslackdir:tofu/slackworkspace:defaultgoogle_playdir:tofu/google_playworkspace:defaultgitdir:tofu/gitworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultkeycloakdir:tofu/keycloakworkspace:default1. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #207. To continue, delete the lock from #207 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #207. To continue, delete the lock from #207 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
google_playdir:tofu/google_playworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #207. To continue, delete the lock from #207 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #207. To continue, delete the lock from #207 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #207. To continue, delete the lock from #207 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.6. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #207. To continue, delete the lock from #207 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.Plan Summary
6 projects, 0 with changes, 0 with no changes, 6 failed
Are you also joining CMU Housing? Just checking
@krishsax wrote in #270 (comment):
I contributed to CMU Housing over the summer so I already joined, but I would like to also join slai
superseded by #422
Pull request closed