feat: add qianxuege to slai #189
No reviewers
Labels
No labels
needs/docs
needs/research
needs/upstream
type/bug
type/external
type/feature
type/refactor
urgency
high
urgency
immediate
urgency
low
urgency
medium
No milestone
No project
No assignees
3 participants
Notifications
Due date
No due date set.
Dependencies
No dependencies set
Reference
ScottyLabs/governance!189
Loading…
Reference in a new issue
No description provided.
Delete branch "qianxuege/governance:main"
Deleting a branch is permanent. Although the deleted branch may continue to exist for a short time before it actually gets removed, it CANNOT be undone in most cases. Continue?
@qianxuege wrote in #189 (comment):
@qianxuege
see step 1 in the readme
@anish
just linked discord
Ran Plan for 6 projects:
slackdir:tofu/slackworkspace:defaultgitdir:tofu/gitworkspace:defaultgoogle_playdir:tofu/google_playworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultdiscorddir:tofu/discordworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #188. To continue, delete the lock from #188 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #188. To continue, delete the lock from #188 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
google_playdir:tofu/google_playworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #188. To continue, delete the lock from #188 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #188. To continue, delete the lock from #188 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #188. To continue, delete the lock from #188 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.6. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #188. To continue, delete the lock from #188 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.Plan Summary
6 projects, 0 with changes, 0 with no changes, 6 failed
atlantis plan
Ran Plan for 6 projects:
slackdir:tofu/slackworkspace:defaultgoogle_playdir:tofu/google_playworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 4 to add, 0 to change, 3 to destroy.
2. project:
google_playdir:tofu/google_playworkspace:defaultShow Output
Plan: 3 to add, 0 to change, 2 to destroy.
3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 2 to destroy.
4. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 0 to add, 7 to change, 0 to destroy.
5. project:
discorddir:tofu/discordworkspace:defaultShow Output
Plan: 1 to add, 2 to change, 1 to destroy.
6. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 6 to add, 2 to change, 2 to destroy.
Plan Summary
6 projects, 6 with changes, 0 with no changes, 0 failed
atlantis apply
Ran Apply for 6 projects:
slackdir:tofu/slackworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgoogle_playdir:tofu/google_playworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
2. project:
discorddir:tofu/discordworkspace:defaultApply Error
Show Output
3. project:
google_playdir:tofu/google_playworkspace:defaultShow Output
4. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
5. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
6. project:
gitdir:tofu/gitworkspace:defaultShow Output
Apply Summary
6 projects, 5 successful, 0 failed, 1 errored
@qianxuege you have to join the scottylabs discord
atlantis unlock
All Atlantis locks for this PR have been unlocked and plans discarded
6134811bca4311e2af99Ran Plan for 6 projects:
slackdir:tofu/slackworkspace:defaultgoogle_playdir:tofu/google_playworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 0 to destroy.
2. project:
google_playdir:tofu/google_playworkspace:default3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 0 to destroy.
4. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 0 to add, 5 to change, 0 to destroy.
5. project:
discorddir:tofu/discordworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 0 to destroy.
6. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 3 to add, 1 to change, 0 to destroy.
Plan Summary
6 projects, 5 with changes, 1 with no changes, 0 failed
atlantis apply
Ran Apply for 6 projects:
google_playdir:tofu/google_playworkspace:defaultslackdir:tofu/slackworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultgitdir:tofu/gitworkspace:defaultkeycloakdir:tofu/keycloakworkspace:default1. project:
google_playdir:tofu/google_playworkspace:defaultShow Output
2. project:
slackdir:tofu/slackworkspace:defaultShow Output
3. project:
discorddir:tofu/discordworkspace:defaultShow Output
4. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
5. project:
gitdir:tofu/gitworkspace:defaultShow Output
6. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Apply Summary
6 projects, 6 successful, 0 failed, 0 errored
Locks and plans deleted for the projects and workspaces modified in this pull request:
tofu/discordworkspace:defaulttofu/gitworkspace:defaulttofu/google_groupsworkspace:defaulttofu/google_playworkspace:defaulttofu/keycloakworkspace:defaulttofu/slackworkspace:default