feat: add Acchs to OCC #191

Merged
anish merged 2 commits from acchs/governance:acchs-to-occ into main 2026-09-06 04:05:52 +00:00
Member

fixed integration issue with #190

fixed integration issue with #190
Added Acchs Zhou to OCC
Some checks are pending
kennel/build skipped: no services or static sites defined
CI / check (pull_request) Blocked by required conditions
CI / check-1 (pull_request) Blocked by required conditions
CI / build (pull_request) Blocked by required conditions
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
Validate / Governance (pull_request_target) Successful in 2m52s
86819c2963
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 1 to add, 0 to change, 2 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 1 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 6 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 2 to add, 2 to change, 2 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
fdcf1aebd3

Ran Plan for 6 projects:

  1. project: google_groups dir: tofu/google_groups workspace: default
  2. project: keycloak dir: tofu/keycloak workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: slack dir: tofu/slack workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: google_play dir: tofu/google_play workspace: default

1. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #189. To continue, delete the lock from #189 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #189. To continue, delete the lock from #189 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #189. To continue, delete the lock from #189 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #189. To continue, delete the lock from #189 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #189. To continue, delete the lock from #189 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #189. To continue, delete the lock from #189 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` --- ### 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #189. To continue, delete the lock from #189 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #189. To continue, delete the lock from #189 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #189. To continue, delete the lock from #189 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #189. To continue, delete the lock from #189 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #189. To continue, delete the lock from #189 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #189. To continue, delete the lock from #189 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.31s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_hub_acchs will be created
+ resource "null_resource" "slack_hub_acchs" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0BVBNX8YGZ"
        }
    }

  # null_resource.slack_hub_qianxuege will be destroyed
  # (because null_resource.slack_hub_qianxuege is not in configuration)
- resource "null_resource" "slack_hub_qianxuege" {
-       id       = "86044828863040854" -> null
-       triggers = {
          - "channel" = "C096TM8EMS8"
          - "user"    = "U05TML2G1N0"
        } -> null
    }

  # null_resource.slack_slai_ch0_qianxuege will be destroyed
  # (because null_resource.slack_slai_ch0_qianxuege is not in configuration)
- resource "null_resource" "slack_slai_ch0_qianxuege" {
-       id       = "1393997705975323500" -> null
-       triggers = {
          - "channel" = "C08SRUGUQ05"
          - "user"    = "U05TML2G1N0"
        } -> null
    }

Plan: 1 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 1 to add, 0 to change, 2 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.29s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.27s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_acchs will be created
+ resource "google_cloud_identity_group_membership" "tech_acchs" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "azzhou@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

  # google_cloud_identity_group_membership.tech_qianxuege will be destroyed
  # (because google_cloud_identity_group_membership.tech_qianxuege is not in configuration)
- resource "google_cloud_identity_group_membership" "tech_qianxuege" {
-       create_time = "2026-09-06T03:07:00.351228Z" -> null
-       group       = "groups/04anzqyu1bqbv79" -> null
-       id          = "groups/04anzqyu1bqbv79/memberships/107376351055066347602" -> null
-       name        = "groups/04anzqyu1bqbv79/memberships/107376351055066347602" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-06T03:07:00.351228Z" -> null

-       preferred_member_key {
-           id = "qge@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

Plan: 1 to add, 0 to change, 1 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 1 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.29s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_occ_occ-main will be updated in-place
! resource "keycloak_group_memberships" "project_occ_occ-main" {
        id       = "scottylabs/group-memberships/a4e98eec-77cf-49f8-8ea6-32cf6332cf36"
!       members  = [
+           "azzhou",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-agent will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-agent" {
        id       = "scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423"
!       members  = [
-           "qge",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-mcp-server will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server" {
        id       = "scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8"
!       members  = [
-           "qge",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-sms-surface will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface" {
        id       = "scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43"
!       members  = [
-           "qge",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-surface will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-surface" {
        id       = "scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775"
!       members  = [
-           "qge",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
-           "qge",
+           "azzhou",
            # (28 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 6 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 6 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
   Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/191/default/crates/governance-schema)
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/191/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/191/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/191/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 9.23s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.acchs will be created
+ resource "discord_member_roles" "acchs" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "781930226778767380"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1543486517908082728"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.23s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)
- destroy

OpenTofu will perform the following actions:

  # forgejo_team_member.occ_acchs will be created
+ resource "forgejo_team_member" "occ_acchs" {
+       team_id = 33
+       user    = "acchs"
    }

  # forgejo_team_member.slai_qianxuege will be destroyed
  # (because forgejo_team_member.slai_qianxuege is not in configuration)
- resource "forgejo_team_member" "slai_qianxuege" {
-       team_id = 8 -> null
-       user    = "qianxuege" -> null
    }

  # github_repository.slai_cmugpt_surface will be updated in-place
! resource "github_repository" "slai_cmugpt_surface" {
        id                                      = "cmugpt-surface"
        name                                    = "cmugpt-surface"
        # (39 unchanged attributes hidden)

-       template {
-           include_all_branches = false -> null
-           owner                = "scottylabs-labrador" -> null
-           repository           = "ScottyStack" -> null
        }

        # (1 unchanged block hidden)
    }

  # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place
! resource "github_repository_ruleset" "cmu-eats_eats_ruleset" {
        id          = "21330218"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 161775995
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_team_membership.occ_acchs will be created
+ resource "github_team_membership" "occ_acchs" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "19240310"
+       username = "acchs-at-cmu"
    }

  # github_team_membership.slai_qianxuege will be destroyed
  # (because github_team_membership.slai_qianxuege is not in configuration)
- resource "github_team_membership" "slai_qianxuege" {
-       etag     = "W/\"5592d489c16f66c4be701aecf06ee22b9a09fb571ef6ec232e4fd5fba15b5118\"" -> null
-       id       = "17571328:qianxuege" -> null
-       role     = "member" -> null
-       team_id  = "17571328" -> null
-       username = "qianxuege" -> null
    }

Plan: 2 to add, 2 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 2 to change, 2 to destroy.


Plan Summary

6 projects, 5 with changes, 1 with no changes, 0 failed

  • To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.31s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # null_resource.slack_hub_acchs will be created + resource "null_resource" "slack_hub_acchs" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0BVBNX8YGZ" } } # null_resource.slack_hub_qianxuege will be destroyed # (because null_resource.slack_hub_qianxuege is not in configuration) - resource "null_resource" "slack_hub_qianxuege" { - id = "86044828863040854" -> null - triggers = { - "channel" = "C096TM8EMS8" - "user" = "U05TML2G1N0" } -> null } # null_resource.slack_slai_ch0_qianxuege will be destroyed # (because null_resource.slack_slai_ch0_qianxuege is not in configuration) - resource "null_resource" "slack_slai_ch0_qianxuege" { - id = "1393997705975323500" -> null - triggers = { - "channel" = "C08SRUGUQ05" - "user" = "U05TML2G1N0" } -> null } Plan: 1 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 1 to add, 0 to change, 2 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.29s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.27s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_acchs will be created + resource "google_cloud_identity_group_membership" "tech_acchs" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "azzhou@andrew.cmu.edu" } + roles { + name = "MEMBER" } } # google_cloud_identity_group_membership.tech_qianxuege will be destroyed # (because google_cloud_identity_group_membership.tech_qianxuege is not in configuration) - resource "google_cloud_identity_group_membership" "tech_qianxuege" { - create_time = "2026-09-06T03:07:00.351228Z" -> null - group = "groups/04anzqyu1bqbv79" -> null - id = "groups/04anzqyu1bqbv79/memberships/107376351055066347602" -> null - name = "groups/04anzqyu1bqbv79/memberships/107376351055066347602" -> null - type = "USER" -> null - update_time = "2026-09-06T03:07:00.351228Z" -> null - preferred_member_key { - id = "qge@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } Plan: 1 to add, 0 to change, 1 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 1 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.29s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_occ_occ-main will be updated in-place ! resource "keycloak_group_memberships" "project_occ_occ-main" { id = "scottylabs/group-memberships/a4e98eec-77cf-49f8-8ea6-32cf6332cf36" ! members = [ + "azzhou", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-agent will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-agent" { id = "scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423" ! members = [ - "qge", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-mcp-server will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server" { id = "scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8" ! members = [ - "qge", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-sms-surface will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface" { id = "scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43" ! members = [ - "qge", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-surface will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-surface" { id = "scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775" ! members = [ - "qge", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ - "qge", + "azzhou", # (28 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 6 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 6 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/191/default/crates/governance-schema) Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/191/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/191/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/191/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 9.23s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.acchs will be created + resource "discord_member_roles" "acchs" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "781930226778767380" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1543486517908082728" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.23s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) - destroy OpenTofu will perform the following actions: # forgejo_team_member.occ_acchs will be created + resource "forgejo_team_member" "occ_acchs" { + team_id = 33 + user = "acchs" } # forgejo_team_member.slai_qianxuege will be destroyed # (because forgejo_team_member.slai_qianxuege is not in configuration) - resource "forgejo_team_member" "slai_qianxuege" { - team_id = 8 -> null - user = "qianxuege" -> null } # github_repository.slai_cmugpt_surface will be updated in-place ! resource "github_repository" "slai_cmugpt_surface" { id = "cmugpt-surface" name = "cmugpt-surface" # (39 unchanged attributes hidden) - template { - include_all_branches = false -> null - owner = "scottylabs-labrador" -> null - repository = "ScottyStack" -> null } # (1 unchanged block hidden) } # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place ! resource "github_repository_ruleset" "cmu-eats_eats_ruleset" { id = "21330218" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 161775995 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_team_membership.occ_acchs will be created + resource "github_team_membership" "occ_acchs" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "19240310" + username = "acchs-at-cmu" } # github_team_membership.slai_qianxuege will be destroyed # (because github_team_membership.slai_qianxuege is not in configuration) - resource "github_team_membership" "slai_qianxuege" { - etag = "W/\"5592d489c16f66c4be701aecf06ee22b9a09fb571ef6ec232e4fd5fba15b5118\"" -> null - id = "17571328:qianxuege" -> null - role = "member" -> null - team_id = "17571328" -> null - username = "qianxuege" -> null } Plan: 2 to add, 2 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 2 to change, 2 to destroy. --- ### Plan Summary 6 projects, 5 with changes, 1 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis apply

atlantis apply

Ran Apply for 6 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.20s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

2. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.36s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_hub_qianxuege: Destroying... [id=86044828863040854]
null_resource.slack_slai_ch0_qianxuege: Destroying... [id=1393997705975323500]
null_resource.slack_hub_qianxuege: Destruction complete after 0s
null_resource.slack_slai_ch0_qianxuege: Destruction complete after 0s
null_resource.slack_hub_acchs: Creating...
null_resource.slack_hub_acchs: Provisioning with 'local-exec'...
null_resource.slack_hub_acchs (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0BVBNX8YGZ"]
null_resource.slack_hub_acchs: Creation complete after 0s [id=5543954889836015881]

Apply complete! Resources: 1 added, 0 changed, 2 destroyed.

3. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.33s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.acchs: Creating...
discord_member_roles.acchs: Creation complete after 1s [id=1353118209024331948:781930226778767380]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.27s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_qianxuege: Destroying... [id=groups/04anzqyu1bqbv79/memberships/107376351055066347602]
google_cloud_identity_group_membership.tech_acchs: Creating...
google_cloud_identity_group_membership.tech_qianxuege: Destruction complete after 1s
google_cloud_identity_group_membership.tech_acchs: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/102039638856706939215]

Apply complete! Resources: 1 added, 0 changed, 1 destroyed.

5. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.30s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_slai_cmugpt-agent: Modifying... [id=scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423]
keycloak_group_memberships.project_slai_cmugpt-surface: Modifying... [id=scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775]
keycloak_group_memberships.project_occ_occ-main: Modifying... [id=scottylabs/group-memberships/a4e98eec-77cf-49f8-8ea6-32cf6332cf36]
keycloak_group_memberships.project_slai_cmugpt-sms-surface: Modifying... [id=scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43]
keycloak_group_memberships.project_slai_cmugpt-mcp-server: Modifying... [id=scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_slai_cmugpt-mcp-server: Modifications complete after 1s [id=scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8]
keycloak_group_memberships.project_slai_cmugpt-surface: Modifications complete after 1s [id=scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775]
keycloak_group_memberships.project_slai_cmugpt-sms-surface: Modifications complete after 1s [id=scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43]
keycloak_group_memberships.project_slai_cmugpt-agent: Modifications complete after 1s [id=scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423]
keycloak_group_memberships.project_occ_occ-main: Modifications complete after 4s [id=scottylabs/group-memberships/a4e98eec-77cf-49f8-8ea6-32cf6332cf36]
keycloak_group_memberships.projects: Modifications complete after 4s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 6 changed, 0 destroyed.

6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.19s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.slai_qianxuege: Destroying...
forgejo_team_member.occ_acchs: Creating...
forgejo_team_member.occ_acchs: Creation complete after 0s
forgejo_team_member.slai_qianxuege: Destruction complete after 0s
github_team_membership.slai_qianxuege: Destroying... [id=17571328:qianxuege]
github_team_membership.occ_acchs: Creating...
github_repository.slai_cmugpt_surface: Modifying... [id=cmugpt-surface]
github_repository_ruleset.cmu-eats_eats_ruleset: Modifying... [id=21330218]
github_team_membership.slai_qianxuege: Destruction complete after 0s
github_team_membership.occ_acchs: Creation complete after 4s [id=19240310:acchs-at-cmu]
github_repository.slai_cmugpt_surface: Modifications complete after 4s [id=cmugpt-surface]
github_repository_ruleset.cmu-eats_eats_ruleset: Modifications complete after 5s [id=21330218]

Apply complete! Resources: 2 added, 2 changed, 2 destroyed.

Apply Summary

6 projects, 6 successful, 0 failed, 0 errored

Ran Apply for 6 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.20s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.36s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_hub_qianxuege: Destroying... [id=86044828863040854] null_resource.slack_slai_ch0_qianxuege: Destroying... [id=1393997705975323500] null_resource.slack_hub_qianxuege: Destruction complete after 0s null_resource.slack_slai_ch0_qianxuege: Destruction complete after 0s null_resource.slack_hub_acchs: Creating... null_resource.slack_hub_acchs: Provisioning with 'local-exec'... null_resource.slack_hub_acchs (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0BVBNX8YGZ"] null_resource.slack_hub_acchs: Creation complete after 0s [id=5543954889836015881] Apply complete! Resources: 1 added, 0 changed, 2 destroyed. ``` </details> --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.33s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.acchs: Creating... discord_member_roles.acchs: Creation complete after 1s [id=1353118209024331948:781930226778767380] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.27s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_qianxuege: Destroying... [id=groups/04anzqyu1bqbv79/memberships/107376351055066347602] google_cloud_identity_group_membership.tech_acchs: Creating... google_cloud_identity_group_membership.tech_qianxuege: Destruction complete after 1s google_cloud_identity_group_membership.tech_acchs: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/102039638856706939215] Apply complete! Resources: 1 added, 0 changed, 1 destroyed. ``` </details> --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.30s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_slai_cmugpt-agent: Modifying... [id=scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423] keycloak_group_memberships.project_slai_cmugpt-surface: Modifying... [id=scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775] keycloak_group_memberships.project_occ_occ-main: Modifying... [id=scottylabs/group-memberships/a4e98eec-77cf-49f8-8ea6-32cf6332cf36] keycloak_group_memberships.project_slai_cmugpt-sms-surface: Modifying... [id=scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43] keycloak_group_memberships.project_slai_cmugpt-mcp-server: Modifying... [id=scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_slai_cmugpt-mcp-server: Modifications complete after 1s [id=scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8] keycloak_group_memberships.project_slai_cmugpt-surface: Modifications complete after 1s [id=scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775] keycloak_group_memberships.project_slai_cmugpt-sms-surface: Modifications complete after 1s [id=scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43] keycloak_group_memberships.project_slai_cmugpt-agent: Modifications complete after 1s [id=scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423] keycloak_group_memberships.project_occ_occ-main: Modifications complete after 4s [id=scottylabs/group-memberships/a4e98eec-77cf-49f8-8ea6-32cf6332cf36] keycloak_group_memberships.projects: Modifications complete after 4s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 6 changed, 0 destroyed. ``` </details> --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.19s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.slai_qianxuege: Destroying... forgejo_team_member.occ_acchs: Creating... forgejo_team_member.occ_acchs: Creation complete after 0s forgejo_team_member.slai_qianxuege: Destruction complete after 0s github_team_membership.slai_qianxuege: Destroying... [id=17571328:qianxuege] github_team_membership.occ_acchs: Creating... github_repository.slai_cmugpt_surface: Modifying... [id=cmugpt-surface] github_repository_ruleset.cmu-eats_eats_ruleset: Modifying... [id=21330218] github_team_membership.slai_qianxuege: Destruction complete after 0s github_team_membership.occ_acchs: Creation complete after 4s [id=19240310:acchs-at-cmu] github_repository.slai_cmugpt_surface: Modifications complete after 4s [id=cmugpt-surface] github_repository_ruleset.cmu-eats_eats_ruleset: Modifications complete after 5s [id=21330218] Apply complete! Resources: 2 added, 2 changed, 2 destroyed. ``` </details> --- ### Apply Summary 6 projects, 6 successful, 0 failed, 0 errored
anish merged commit ba55b8f6c8 into main 2026-09-06 04:05:52 +00:00
anish deleted branch acchs-to-occ 2026-09-06 04:05:52 +00:00

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
anish referenced this pull request from a commit 2026-09-06 04:05:53 +00:00
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 1 to add, 0 to change, 2 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 1 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 6 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 2 to add, 2 to change, 2 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
Some workflows are waiting to be reviewed.
Sign in to join this conversation.
No description provided.