feat: add siddharthr as lead to CMU Maps #188

Merged
thesuperRL merged 3 commits from siddharthr/governance:main into main 2026-09-06 15:35:38 +00:00
Member
No description provided.
feat: add siddharthr as lead to CMU Maps
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Failing after 4m19s
CI / build (pull_request) Has been skipped
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
8e0f5d1f3b
chore: remove temp leads
All checks were successful
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
CI / check-1 (pull_request) Successful in 5m52s
CI / check (pull_request) Successful in 0s
Validate / Governance (pull_request_target) Successful in 3m49s
cdc8443fd7
Signed-off-by: anish <i@anish.land>
Owner

@siddharthr can you link your accounts (see the link in step 1 in the readme)

error: user siddharthr has not linked their slack account in keycloak

the only one you don't have to link is codeberg

@siddharthr can you link your accounts (see the link in step 1 in the readme) ``` error: user siddharthr has not linked their slack account in keycloak ``` the only one you don't have to link is codeberg
Author
Member

Went ahead and linked missing accounts

Went ahead and linked missing accounts
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/apply: git Apply failed.
atlantis/apply: slack Apply succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 1/6 projects failed to apply.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: slack Plan: 3 to add, 0 to change, 3 to destroy.
atlantis/plan: google_play Plan: 2 to add, 0 to change, 3 to destroy.
atlantis/plan: google_groups Plan: 2 to add, 0 to change, 1 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 4 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 2 to change, 1 to destroy.
atlantis/plan: git Plan failed.
atlantis/plan 1/6 projects failed to plan.
02499a9049

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.55s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-maps_ch0_krishsax will be destroyed
  # (because null_resource.slack_cmu-maps_ch0_krishsax is not in configuration)
- resource "null_resource" "slack_cmu-maps_ch0_krishsax" {
-       id       = "1435654462017329264" -> null
-       triggers = {
          - "channel" = "C05T7QRN4LF"
          - "user"    = "U09DL6NH8FL"
        } -> null
    }

  # null_resource.slack_cmu-maps_ch0_siddharthr will be created
+ resource "null_resource" "slack_cmu-maps_ch0_siddharthr" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C05T7QRN4LF"
          + "user"    = "U09DC6MFB5E"
        }
    }

  # null_resource.slack_cmu-maps_ch0_xboxbedrock will be destroyed
  # (because null_resource.slack_cmu-maps_ch0_xboxbedrock is not in configuration)
- resource "null_resource" "slack_cmu-maps_ch0_xboxbedrock" {
-       id       = "5341413456122565893" -> null
-       triggers = {
          - "channel" = "C05T7QRN4LF"
          - "user"    = "U0AQRKYLGHK"
        } -> null
    }

  # null_resource.slack_hub_siddharthr will be created
+ resource "null_resource" "slack_hub_siddharthr" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U09DC6MFB5E"
        }
    }

  # null_resource.slack_leads_siddharthr will be created
+ resource "null_resource" "slack_leads_siddharthr" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0AN1M0HDV0"
          + "user"    = "U09DC6MFB5E"
        }
    }

Plan: 3 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 3 to add, 0 to change, 2 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.51s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # googleplay_app_iam.maps_krishsax will be destroyed
  # (because googleplay_app_iam.maps_krishsax is not in configuration)
- resource "googleplay_app_iam" "maps_krishsax" {
-       app_id               = "com.cmumaps.app" -> null
-       expanded_permissions = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       permissions          = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       user_id              = "krishsax@andrew.cmu.edu" -> null
    }

  # googleplay_app_iam.maps_siddharthr will be created
+ resource "googleplay_app_iam" "maps_siddharthr" {
+       app_id               = "com.cmumaps.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_APP_CONTENT",
+           "CAN_MANAGE_PUBLIC_APKS",
+           "CAN_MANAGE_PUBLIC_LISTING",
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_REPLY_TO_REVIEWS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "sradhak2@andrew.cmu.edu"
    }

  # googleplay_app_iam.maps_xboxbedrock will be destroyed
  # (because googleplay_app_iam.maps_xboxbedrock is not in configuration)
- resource "googleplay_app_iam" "maps_xboxbedrock" {
-       app_id               = "com.cmumaps.app" -> null
-       expanded_permissions = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       permissions          = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       user_id              = "spola2@andrew.cmu.edu" -> null
    }

  # googleplay_user.krishsax will be destroyed
  # (because googleplay_user.krishsax is not in configuration)
- resource "googleplay_user" "krishsax" {
-       email                = "krishsax@andrew.cmu.edu" -> null
-       expanded_permissions = [
-           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ] -> null
-       global_permissions   = [
-           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ] -> null
-       id                   = "krishsax@andrew.cmu.edu" -> null
-       name                 = "developers/7984360456919420393/users/krishsax@andrew.cmu.edu" -> null
    }

  # googleplay_user.siddharthr will be created
+ resource "googleplay_user" "siddharthr" {
+       email                = "sradhak2@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

Plan: 2 to add, 0 to change, 3 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 2 to add, 0 to change, 3 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.50s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.ops_siddharthr will be created
+ resource "google_cloud_identity_group_membership" "ops_siddharthr" {
+       create_time = (known after apply)
+       group       = "groups/03fwokq036zsp1f"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "sradhak2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

  # google_cloud_identity_group_membership.tech_siddharthr will be created
+ resource "google_cloud_identity_group_membership" "tech_siddharthr" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "sradhak2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 2 to add, 0 to change, 0 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
   Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/188/default/crates/governance-schema)
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/188/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/188/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/188/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 9.44s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-maps_maps will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-maps_maps" {
        id       = "scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64"
!       members  = [
-           "krishsax",
-           "spola2",
+           "sradhak2",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-maps_maps_admins will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-maps_maps_admins" {
        id       = "scottylabs/group-memberships/75db0c7f-7cfb-4d89-b84f-0f50523ed55b"
!       members  = [
-           "krishsax",
-           "spola2",
+           "sradhak2",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "sradhak2",
            # (28 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 3 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 3 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.49s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # discord_member_roles.krishsax will be updated in-place
! resource "discord_member_roles" "krishsax" {
        id        = "1353118209024331948:782966391321395240"
        # (2 unchanged attributes hidden)

-       role {
-           has_role = true -> null
-           role_id  = "1526003694708195448" -> null
        }

        # (3 unchanged blocks hidden)
    }

  # discord_member_roles.siddharthr will be created
+ resource "discord_member_roles" "siddharthr" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "584841698186756167"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1519792320755662928"
        }
+       role {
+           has_role = true
+           role_id  = "1526003694708195448"
        }
    }

  # discord_member_roles.xboxbedrock will be updated in-place
! resource "discord_member_roles" "xboxbedrock" {
        id        = "1353118209024331948:712398210345205871"
        # (2 unchanged attributes hidden)

-       role {
-           has_role = true -> null
-           role_id  = "1526003694708195448" -> null
        }

        # (5 unchanged blocks hidden)
    }

Plan: 1 to add, 2 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 2 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.56s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)
- destroy

OpenTofu will perform the following actions:

  # forgejo_team_member.cmu-maps_krishsax will be destroyed
  # (because forgejo_team_member.cmu-maps_krishsax is not in configuration)
- resource "forgejo_team_member" "cmu-maps_krishsax" {
-       team_id = 13 -> null
-       user    = "krishsax" -> null
    }

  # forgejo_team_member.cmu-maps_siddharthr will be created
+ resource "forgejo_team_member" "cmu-maps_siddharthr" {
+       team_id = 13
+       user    = "sradhak2"
    }

  # forgejo_team_member.cmu-maps_xboxbedrock will be destroyed
  # (because forgejo_team_member.cmu-maps_xboxbedrock is not in configuration)
- resource "forgejo_team_member" "cmu-maps_xboxbedrock" {
-       team_id = 13 -> null
-       user    = "xboxbedrock" -> null
    }

  # forgejo_team_member.maps_leads_krishsax will be destroyed
  # (because forgejo_team_member.maps_leads_krishsax is not in configuration)
- resource "forgejo_team_member" "maps_leads_krishsax" {
-       team_id = 38 -> null
-       user    = "krishsax" -> null
    }

  # forgejo_team_member.maps_leads_siddharthr will be created
+ resource "forgejo_team_member" "maps_leads_siddharthr" {
+       team_id = 38
+       user    = "sradhak2"
    }

  # forgejo_team_member.maps_leads_xboxbedrock will be destroyed
  # (because forgejo_team_member.maps_leads_xboxbedrock is not in configuration)
- resource "forgejo_team_member" "maps_leads_xboxbedrock" {
-       team_id = 38 -> null
-       user    = "xboxbedrock" -> null
    }

  # github_repository.slai_cmugpt_surface will be updated in-place
! resource "github_repository" "slai_cmugpt_surface" {
        id                                      = "cmugpt-surface"
        name                                    = "cmugpt-surface"
        # (39 unchanged attributes hidden)

-       template {
-           include_all_branches = false -> null
-           owner                = "scottylabs-labrador" -> null
-           repository           = "ScottyStack" -> null
        }

        # (1 unchanged block hidden)
    }

  # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place
! resource "github_repository_ruleset" "cmu-eats_eats_ruleset" {
        id          = "21330218"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 161775995
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_team_membership.cmu-maps_krishsax will be destroyed
  # (because github_team_membership.cmu-maps_krishsax is not in configuration)
- resource "github_team_membership" "cmu-maps_krishsax" {
-       etag     = "W/\"0ca81c85b299f01891dc704fc1e1f11d9b5e73a022c43d28f0593b116aa3098f\"" -> null
-       id       = "10872360:krishsaxena" -> null
-       role     = "maintainer" -> null
-       team_id  = "10872360" -> null
-       username = "krishsaxena" -> null
    }

  # github_team_membership.cmu-maps_siddharthr will be created
+ resource "github_team_membership" "cmu-maps_siddharthr" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "maintainer"
+       team_id  = "10872360"
+       username = "overthrowing"
    }

  # github_team_membership.cmu-maps_xboxbedrock will be destroyed
  # (because github_team_membership.cmu-maps_xboxbedrock is not in configuration)
- resource "github_team_membership" "cmu-maps_xboxbedrock" {
-       etag     = "W/\"695105377afc546ab133d841be4a870c042dca48a6172b3388875744e9e1aef9\"" -> null
-       id       = "10872360:xboxbedrock" -> null
-       role     = "maintainer" -> null
-       team_id  = "10872360" -> null
-       username = "xboxbedrock" -> null
    }

Plan: 3 to add, 2 to change, 6 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 3 to add, 2 to change, 6 to destroy.


Plan Summary

6 projects, 6 with changes, 0 with no changes, 0 failed

  • To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.55s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # null_resource.slack_cmu-maps_ch0_krishsax will be destroyed # (because null_resource.slack_cmu-maps_ch0_krishsax is not in configuration) - resource "null_resource" "slack_cmu-maps_ch0_krishsax" { - id = "1435654462017329264" -> null - triggers = { - "channel" = "C05T7QRN4LF" - "user" = "U09DL6NH8FL" } -> null } # null_resource.slack_cmu-maps_ch0_siddharthr will be created + resource "null_resource" "slack_cmu-maps_ch0_siddharthr" { + id = (known after apply) + triggers = { + "channel" = "C05T7QRN4LF" + "user" = "U09DC6MFB5E" } } # null_resource.slack_cmu-maps_ch0_xboxbedrock will be destroyed # (because null_resource.slack_cmu-maps_ch0_xboxbedrock is not in configuration) - resource "null_resource" "slack_cmu-maps_ch0_xboxbedrock" { - id = "5341413456122565893" -> null - triggers = { - "channel" = "C05T7QRN4LF" - "user" = "U0AQRKYLGHK" } -> null } # null_resource.slack_hub_siddharthr will be created + resource "null_resource" "slack_hub_siddharthr" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U09DC6MFB5E" } } # null_resource.slack_leads_siddharthr will be created + resource "null_resource" "slack_leads_siddharthr" { + id = (known after apply) + triggers = { + "channel" = "C0AN1M0HDV0" + "user" = "U09DC6MFB5E" } } Plan: 3 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 3 to add, 0 to change, 2 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.51s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # googleplay_app_iam.maps_krishsax will be destroyed # (because googleplay_app_iam.maps_krishsax is not in configuration) - resource "googleplay_app_iam" "maps_krishsax" { - app_id = "com.cmumaps.app" -> null - expanded_permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - user_id = "krishsax@andrew.cmu.edu" -> null } # googleplay_app_iam.maps_siddharthr will be created + resource "googleplay_app_iam" "maps_siddharthr" { + app_id = "com.cmumaps.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_APP_CONTENT", + "CAN_MANAGE_PUBLIC_APKS", + "CAN_MANAGE_PUBLIC_LISTING", + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_REPLY_TO_REVIEWS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "sradhak2@andrew.cmu.edu" } # googleplay_app_iam.maps_xboxbedrock will be destroyed # (because googleplay_app_iam.maps_xboxbedrock is not in configuration) - resource "googleplay_app_iam" "maps_xboxbedrock" { - app_id = "com.cmumaps.app" -> null - expanded_permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - user_id = "spola2@andrew.cmu.edu" -> null } # googleplay_user.krishsax will be destroyed # (because googleplay_user.krishsax is not in configuration) - resource "googleplay_user" "krishsax" { - email = "krishsax@andrew.cmu.edu" -> null - expanded_permissions = [ - "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] -> null - global_permissions = [ - "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] -> null - id = "krishsax@andrew.cmu.edu" -> null - name = "developers/7984360456919420393/users/krishsax@andrew.cmu.edu" -> null } # googleplay_user.siddharthr will be created + resource "googleplay_user" "siddharthr" { + email = "sradhak2@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } Plan: 2 to add, 0 to change, 3 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 2 to add, 0 to change, 3 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.50s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.ops_siddharthr will be created + resource "google_cloud_identity_group_membership" "ops_siddharthr" { + create_time = (known after apply) + group = "groups/03fwokq036zsp1f" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "sradhak2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } # google_cloud_identity_group_membership.tech_siddharthr will be created + resource "google_cloud_identity_group_membership" "tech_siddharthr" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "sradhak2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/188/default/crates/governance-schema) Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/188/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/188/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/188/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 9.44s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-maps_maps will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-maps_maps" { id = "scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64" ! members = [ - "krishsax", - "spola2", + "sradhak2", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-maps_maps_admins will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-maps_maps_admins" { id = "scottylabs/group-memberships/75db0c7f-7cfb-4d89-b84f-0f50523ed55b" ! members = [ - "krishsax", - "spola2", + "sradhak2", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "sradhak2", # (28 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 3 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 3 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.49s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # discord_member_roles.krishsax will be updated in-place ! resource "discord_member_roles" "krishsax" { id = "1353118209024331948:782966391321395240" # (2 unchanged attributes hidden) - role { - has_role = true -> null - role_id = "1526003694708195448" -> null } # (3 unchanged blocks hidden) } # discord_member_roles.siddharthr will be created + resource "discord_member_roles" "siddharthr" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "584841698186756167" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1519792320755662928" } + role { + has_role = true + role_id = "1526003694708195448" } } # discord_member_roles.xboxbedrock will be updated in-place ! resource "discord_member_roles" "xboxbedrock" { id = "1353118209024331948:712398210345205871" # (2 unchanged attributes hidden) - role { - has_role = true -> null - role_id = "1526003694708195448" -> null } # (5 unchanged blocks hidden) } Plan: 1 to add, 2 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 2 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.56s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) - destroy OpenTofu will perform the following actions: # forgejo_team_member.cmu-maps_krishsax will be destroyed # (because forgejo_team_member.cmu-maps_krishsax is not in configuration) - resource "forgejo_team_member" "cmu-maps_krishsax" { - team_id = 13 -> null - user = "krishsax" -> null } # forgejo_team_member.cmu-maps_siddharthr will be created + resource "forgejo_team_member" "cmu-maps_siddharthr" { + team_id = 13 + user = "sradhak2" } # forgejo_team_member.cmu-maps_xboxbedrock will be destroyed # (because forgejo_team_member.cmu-maps_xboxbedrock is not in configuration) - resource "forgejo_team_member" "cmu-maps_xboxbedrock" { - team_id = 13 -> null - user = "xboxbedrock" -> null } # forgejo_team_member.maps_leads_krishsax will be destroyed # (because forgejo_team_member.maps_leads_krishsax is not in configuration) - resource "forgejo_team_member" "maps_leads_krishsax" { - team_id = 38 -> null - user = "krishsax" -> null } # forgejo_team_member.maps_leads_siddharthr will be created + resource "forgejo_team_member" "maps_leads_siddharthr" { + team_id = 38 + user = "sradhak2" } # forgejo_team_member.maps_leads_xboxbedrock will be destroyed # (because forgejo_team_member.maps_leads_xboxbedrock is not in configuration) - resource "forgejo_team_member" "maps_leads_xboxbedrock" { - team_id = 38 -> null - user = "xboxbedrock" -> null } # github_repository.slai_cmugpt_surface will be updated in-place ! resource "github_repository" "slai_cmugpt_surface" { id = "cmugpt-surface" name = "cmugpt-surface" # (39 unchanged attributes hidden) - template { - include_all_branches = false -> null - owner = "scottylabs-labrador" -> null - repository = "ScottyStack" -> null } # (1 unchanged block hidden) } # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place ! resource "github_repository_ruleset" "cmu-eats_eats_ruleset" { id = "21330218" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 161775995 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_team_membership.cmu-maps_krishsax will be destroyed # (because github_team_membership.cmu-maps_krishsax is not in configuration) - resource "github_team_membership" "cmu-maps_krishsax" { - etag = "W/\"0ca81c85b299f01891dc704fc1e1f11d9b5e73a022c43d28f0593b116aa3098f\"" -> null - id = "10872360:krishsaxena" -> null - role = "maintainer" -> null - team_id = "10872360" -> null - username = "krishsaxena" -> null } # github_team_membership.cmu-maps_siddharthr will be created + resource "github_team_membership" "cmu-maps_siddharthr" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = "10872360" + username = "overthrowing" } # github_team_membership.cmu-maps_xboxbedrock will be destroyed # (because github_team_membership.cmu-maps_xboxbedrock is not in configuration) - resource "github_team_membership" "cmu-maps_xboxbedrock" { - etag = "W/\"695105377afc546ab133d841be4a870c042dca48a6172b3388875744e9e1aef9\"" -> null - id = "10872360:xboxbedrock" -> null - role = "maintainer" -> null - team_id = "10872360" -> null - username = "xboxbedrock" -> null } Plan: 3 to add, 2 to change, 6 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 3 to add, 2 to change, 6 to destroy. --- ### Plan Summary 6 projects, 6 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis apply

atlantis apply

Ran Apply for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: discord dir: tofu/discord workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_cmu-maps_ch0_xboxbedrock: Destroying... [id=5341413456122565893]
null_resource.slack_cmu-maps_ch0_krishsax: Destroying... [id=1435654462017329264]
null_resource.slack_cmu-maps_ch0_krishsax: Destruction complete after 0s
null_resource.slack_cmu-maps_ch0_xboxbedrock: Destruction complete after 0s
null_resource.slack_leads_siddharthr: Creating...
null_resource.slack_hub_siddharthr: Creating...
null_resource.slack_hub_siddharthr: Provisioning with 'local-exec'...
null_resource.slack_hub_siddharthr (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U09DC6MFB5E"]
null_resource.slack_leads_siddharthr: Provisioning with 'local-exec'...
null_resource.slack_leads_siddharthr (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0AN1M0HDV0 --user U09DC6MFB5E"]
null_resource.slack_cmu-maps_ch0_siddharthr: Creating...
null_resource.slack_cmu-maps_ch0_siddharthr: Provisioning with 'local-exec'...
null_resource.slack_cmu-maps_ch0_siddharthr (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C05T7QRN4LF --user U09DC6MFB5E"]
null_resource.slack_leads_siddharthr: Creation complete after 0s [id=6016165231853930076]
null_resource.slack_hub_siddharthr: Creation complete after 0s [id=5258804225368682447]
null_resource.slack_cmu-maps_ch0_siddharthr: Creation complete after 0s [id=5207087142356789523]

Apply complete! Resources: 3 added, 0 changed, 2 destroyed.

2. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.23s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
googleplay_user.siddharthr: Creating...
googleplay_app_iam.maps_krishsax: Destroying...
googleplay_app_iam.maps_xboxbedrock: Destroying...
googleplay_user.siddharthr: Creation complete after 1s [id=sradhak2@andrew.cmu.edu]
googleplay_app_iam.maps_siddharthr: Creating...
googleplay_app_iam.maps_krishsax: Destruction complete after 1s
googleplay_user.krishsax: Destroying... [id=krishsax@andrew.cmu.edu]
googleplay_app_iam.maps_xboxbedrock: Destruction complete after 1s
googleplay_user.krishsax: Destruction complete after 0s
googleplay_app_iam.maps_siddharthr: Creation complete after 1s

Apply complete! Resources: 2 added, 0 changed, 3 destroyed.

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.38s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.ops_siddharthr: Creating...
google_cloud_identity_group_membership.tech_siddharthr: Creating...
google_cloud_identity_group_membership.ops_siddharthr: Creation complete after 1s [id=groups/03fwokq036zsp1f/memberships/117208415664072816789]
google_cloud_identity_group_membership.tech_siddharthr: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/117208415664072816789]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

4. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.31s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.siddharthr: Creating...
discord_member_roles.krishsax: Modifying... [id=1353118209024331948:782966391321395240]
discord_member_roles.xboxbedrock: Modifying... [id=1353118209024331948:712398210345205871]
discord_member_roles.xboxbedrock: Modifications complete after 1s [id=1353118209024331948:712398210345205871]
discord_member_roles.krishsax: Modifications complete after 1s [id=1353118209024331948:782966391321395240]
discord_member_roles.siddharthr: Creation complete after 2s [id=1353118209024331948:584841698186756167]

Apply complete! Resources: 1 added, 2 changed, 0 destroyed.

5. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.28s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cmu-maps_maps_admins: Modifying... [id=scottylabs/group-memberships/75db0c7f-7cfb-4d89-b84f-0f50523ed55b]
keycloak_group_memberships.project_cmu-maps_maps: Modifying... [id=scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_cmu-maps_maps: Modifications complete after 3s [id=scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64]
keycloak_group_memberships.project_cmu-maps_maps_admins: Modifications complete after 3s [id=scottylabs/group-memberships/75db0c7f-7cfb-4d89-b84f-0f50523ed55b]
keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 3 changed, 0 destroyed.

6. project: git dir: tofu/git workspace: default

Apply Error

Show Output
running '/nix/store/v8n4x020rdfr57a0s3xdq46a3mdd5bk7-opentofu-1.12.5/bin/tofu apply -input=false /var/lib/atlantis/repos/ScottyLabs/governance/188/default/tofu/git/git-default.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/188/default/tofu/git': exit status 1
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.35s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.maps_leads_xboxbedrock: Destroying...
forgejo_team_member.maps_leads_krishsax: Destroying...
forgejo_team_member.cmu-maps_krishsax: Destroying...
forgejo_team_member.cmu-maps_xboxbedrock: Destroying...
forgejo_team_member.maps_leads_siddharthr: Creating...
forgejo_team_member.cmu-maps_siddharthr: Creating...
forgejo_team_member.maps_leads_xboxbedrock: Destruction complete after 0s
forgejo_team_member.maps_leads_krishsax: Destruction complete after 0s
forgejo_team_member.maps_leads_siddharthr: Creation complete after 0s
github_team_membership.cmu-maps_xboxbedrock: Destroying... [id=10872360:xboxbedrock]
github_team_membership.cmu-maps_krishsax: Destroying... [id=10872360:krishsaxena]
github_team_membership.cmu-maps_siddharthr: Creating...
github_repository.slai_cmugpt_surface: Modifying... [id=cmugpt-surface]
github_repository_ruleset.cmu-eats_eats_ruleset: Modifying... [id=21330218]
github_team_membership.cmu-maps_krishsax: Destruction complete after 0s
github_team_membership.cmu-maps_xboxbedrock: Destruction complete after 2s
github_team_membership.cmu-maps_siddharthr: Creation complete after 6s [id=10872360:overthrowing]
github_repository_ruleset.cmu-eats_eats_ruleset: Modifications complete after 6s [id=21330218]
github_repository.slai_cmugpt_surface: Modifications complete after 7s [id=cmugpt-surface]
╷
│ Error: Unable to delete team member
│ 
│ Unknown error (status 500): 
╵
╷
│ Error: Unable to delete team member
│ 
│ Unknown error (status 500): 
╵
╷
│ Error: Unable to create team member
│ 
│   with forgejo_team_member.cmu-maps_siddharthr,
│   on forgejo_memberships.tf.json line 103, in resource.forgejo_team_member.cmu-maps_siddharthr:
│  103:       },
│ 
│ Unknown error (status 500): 
╵


Apply Summary

6 projects, 5 successful, 0 failed, 1 errored

Ran Apply for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_cmu-maps_ch0_xboxbedrock: Destroying... [id=5341413456122565893] null_resource.slack_cmu-maps_ch0_krishsax: Destroying... [id=1435654462017329264] null_resource.slack_cmu-maps_ch0_krishsax: Destruction complete after 0s null_resource.slack_cmu-maps_ch0_xboxbedrock: Destruction complete after 0s null_resource.slack_leads_siddharthr: Creating... null_resource.slack_hub_siddharthr: Creating... null_resource.slack_hub_siddharthr: Provisioning with 'local-exec'... null_resource.slack_hub_siddharthr (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U09DC6MFB5E"] null_resource.slack_leads_siddharthr: Provisioning with 'local-exec'... null_resource.slack_leads_siddharthr (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0AN1M0HDV0 --user U09DC6MFB5E"] null_resource.slack_cmu-maps_ch0_siddharthr: Creating... null_resource.slack_cmu-maps_ch0_siddharthr: Provisioning with 'local-exec'... null_resource.slack_cmu-maps_ch0_siddharthr (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C05T7QRN4LF --user U09DC6MFB5E"] null_resource.slack_leads_siddharthr: Creation complete after 0s [id=6016165231853930076] null_resource.slack_hub_siddharthr: Creation complete after 0s [id=5258804225368682447] null_resource.slack_cmu-maps_ch0_siddharthr: Creation complete after 0s [id=5207087142356789523] Apply complete! Resources: 3 added, 0 changed, 2 destroyed. ``` </details> --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.23s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ googleplay_user.siddharthr: Creating... googleplay_app_iam.maps_krishsax: Destroying... googleplay_app_iam.maps_xboxbedrock: Destroying... googleplay_user.siddharthr: Creation complete after 1s [id=sradhak2@andrew.cmu.edu] googleplay_app_iam.maps_siddharthr: Creating... googleplay_app_iam.maps_krishsax: Destruction complete after 1s googleplay_user.krishsax: Destroying... [id=krishsax@andrew.cmu.edu] googleplay_app_iam.maps_xboxbedrock: Destruction complete after 1s googleplay_user.krishsax: Destruction complete after 0s googleplay_app_iam.maps_siddharthr: Creation complete after 1s Apply complete! Resources: 2 added, 0 changed, 3 destroyed. ``` </details> --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.38s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.ops_siddharthr: Creating... google_cloud_identity_group_membership.tech_siddharthr: Creating... google_cloud_identity_group_membership.ops_siddharthr: Creation complete after 1s [id=groups/03fwokq036zsp1f/memberships/117208415664072816789] google_cloud_identity_group_membership.tech_siddharthr: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/117208415664072816789] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.31s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.siddharthr: Creating... discord_member_roles.krishsax: Modifying... [id=1353118209024331948:782966391321395240] discord_member_roles.xboxbedrock: Modifying... [id=1353118209024331948:712398210345205871] discord_member_roles.xboxbedrock: Modifications complete after 1s [id=1353118209024331948:712398210345205871] discord_member_roles.krishsax: Modifications complete after 1s [id=1353118209024331948:782966391321395240] discord_member_roles.siddharthr: Creation complete after 2s [id=1353118209024331948:584841698186756167] Apply complete! Resources: 1 added, 2 changed, 0 destroyed. ``` </details> --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.28s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cmu-maps_maps_admins: Modifying... [id=scottylabs/group-memberships/75db0c7f-7cfb-4d89-b84f-0f50523ed55b] keycloak_group_memberships.project_cmu-maps_maps: Modifying... [id=scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_cmu-maps_maps: Modifications complete after 3s [id=scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64] keycloak_group_memberships.project_cmu-maps_maps_admins: Modifications complete after 3s [id=scottylabs/group-memberships/75db0c7f-7cfb-4d89-b84f-0f50523ed55b] keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 3 changed, 0 destroyed. ``` </details> --- ### 6. project: `git` dir: `tofu/git` workspace: `default` **Apply Error** <details><summary>Show Output</summary> ``` running '/nix/store/v8n4x020rdfr57a0s3xdq46a3mdd5bk7-opentofu-1.12.5/bin/tofu apply -input=false /var/lib/atlantis/repos/ScottyLabs/governance/188/default/tofu/git/git-default.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/188/default/tofu/git': exit status 1 Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.35s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.maps_leads_xboxbedrock: Destroying... forgejo_team_member.maps_leads_krishsax: Destroying... forgejo_team_member.cmu-maps_krishsax: Destroying... forgejo_team_member.cmu-maps_xboxbedrock: Destroying... forgejo_team_member.maps_leads_siddharthr: Creating... forgejo_team_member.cmu-maps_siddharthr: Creating... forgejo_team_member.maps_leads_xboxbedrock: Destruction complete after 0s forgejo_team_member.maps_leads_krishsax: Destruction complete after 0s forgejo_team_member.maps_leads_siddharthr: Creation complete after 0s github_team_membership.cmu-maps_xboxbedrock: Destroying... [id=10872360:xboxbedrock] github_team_membership.cmu-maps_krishsax: Destroying... [id=10872360:krishsaxena] github_team_membership.cmu-maps_siddharthr: Creating... github_repository.slai_cmugpt_surface: Modifying... [id=cmugpt-surface] github_repository_ruleset.cmu-eats_eats_ruleset: Modifying... [id=21330218] github_team_membership.cmu-maps_krishsax: Destruction complete after 0s github_team_membership.cmu-maps_xboxbedrock: Destruction complete after 2s github_team_membership.cmu-maps_siddharthr: Creation complete after 6s [id=10872360:overthrowing] github_repository_ruleset.cmu-eats_eats_ruleset: Modifications complete after 6s [id=21330218] github_repository.slai_cmugpt_surface: Modifications complete after 7s [id=cmugpt-surface] ╷ │ Error: Unable to delete team member │ │ Unknown error (status 500): ╵ ╷ │ Error: Unable to delete team member │ │ Unknown error (status 500): ╵ ╷ │ Error: Unable to create team member │ │ with forgejo_team_member.cmu-maps_siddharthr, │ on forgejo_memberships.tf.json line 103, in resource.forgejo_team_member.cmu-maps_siddharthr: │ 103: }, │ │ Unknown error (status 500): ╵ ``` </details> --- ### Apply Summary 6 projects, 5 successful, 0 failed, 1 errored
Owner

atlantis unlock

atlantis unlock

All Atlantis locks for this PR have been unlocked and plans discarded

All Atlantis locks for this PR have been unlocked and plans discarded
Owner

atlantis plan

atlantis plan
thesuperRL force-pushed main from 02499a9049
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/apply: git Apply failed.
atlantis/apply: slack Apply succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 1/6 projects failed to apply.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: slack Plan: 3 to add, 0 to change, 3 to destroy.
atlantis/plan: google_play Plan: 2 to add, 0 to change, 3 to destroy.
atlantis/plan: google_groups Plan: 2 to add, 0 to change, 1 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 4 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 2 to change, 1 to destroy.
atlantis/plan: git Plan failed.
atlantis/plan 1/6 projects failed to plan.
to 2dc1c13b11
All checks were successful
atlantis/plan 6/6 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 3 to add, 0 to change, 2 to destroy.
atlantis/plan: google_play Plan: 2 to add, 0 to change, 3 to destroy.
atlantis/plan: google_groups Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 3 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 2 to change, 0 to destroy.
atlantis/plan: git Plan: 3 to add, 1 to change, 6 to destroy.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
2026-09-06 15:23:11 +00:00
Compare

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-maps_ch0_krishsax will be destroyed
  # (because null_resource.slack_cmu-maps_ch0_krishsax is not in configuration)
- resource "null_resource" "slack_cmu-maps_ch0_krishsax" {
-       id       = "8867766601843827684" -> null
-       triggers = {
          - "channel" = "C05T7QRN4LF"
          - "user"    = "U09DL6NH8FL"
        } -> null
    }

  # null_resource.slack_cmu-maps_ch0_siddharthr will be created
+ resource "null_resource" "slack_cmu-maps_ch0_siddharthr" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C05T7QRN4LF"
          + "user"    = "U09DC6MFB5E"
        }
    }

  # null_resource.slack_cmu-maps_ch0_xboxbedrock will be destroyed
  # (because null_resource.slack_cmu-maps_ch0_xboxbedrock is not in configuration)
- resource "null_resource" "slack_cmu-maps_ch0_xboxbedrock" {
-       id       = "2901886759953880290" -> null
-       triggers = {
          - "channel" = "C05T7QRN4LF"
          - "user"    = "U0AQRKYLGHK"
        } -> null
    }

  # null_resource.slack_hub_acchs will be destroyed
  # (because null_resource.slack_hub_acchs is not in configuration)
- resource "null_resource" "slack_hub_acchs" {
-       id       = "5543954889836015881" -> null
-       triggers = {
          - "channel" = "C096TM8EMS8"
          - "user"    = "U0BVBNX8YGZ"
        } -> null
    }

  # null_resource.slack_hub_siddharthr will be created
+ resource "null_resource" "slack_hub_siddharthr" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U09DC6MFB5E"
        }
    }

  # null_resource.slack_leads_siddharthr will be created
+ resource "null_resource" "slack_leads_siddharthr" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0AN1M0HDV0"
          + "user"    = "U09DC6MFB5E"
        }
    }

Plan: 3 to add, 0 to change, 3 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 3 to add, 0 to change, 3 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.27s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # googleplay_app_iam.maps_krishsax will be destroyed
  # (because googleplay_app_iam.maps_krishsax is not in configuration)
- resource "googleplay_app_iam" "maps_krishsax" {
-       app_id               = "com.cmumaps.app" -> null
-       expanded_permissions = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       permissions          = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       user_id              = "krishsax@andrew.cmu.edu" -> null
    }

  # googleplay_app_iam.maps_siddharthr will be created
+ resource "googleplay_app_iam" "maps_siddharthr" {
+       app_id               = "com.cmumaps.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_APP_CONTENT",
+           "CAN_MANAGE_PUBLIC_APKS",
+           "CAN_MANAGE_PUBLIC_LISTING",
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_REPLY_TO_REVIEWS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "sradhak2@andrew.cmu.edu"
    }

  # googleplay_app_iam.maps_xboxbedrock will be destroyed
  # (because googleplay_app_iam.maps_xboxbedrock is not in configuration)
- resource "googleplay_app_iam" "maps_xboxbedrock" {
-       app_id               = "com.cmumaps.app" -> null
-       expanded_permissions = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       permissions          = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       user_id              = "spola2@andrew.cmu.edu" -> null
    }

  # googleplay_user.krishsax will be destroyed
  # (because googleplay_user.krishsax is not in configuration)
- resource "googleplay_user" "krishsax" {
-       email                = "krishsax@andrew.cmu.edu" -> null
-       expanded_permissions = [
-           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ] -> null
-       global_permissions   = [
-           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ] -> null
-       id                   = "krishsax@andrew.cmu.edu" -> null
-       name                 = "developers/7984360456919420393/users/krishsax@andrew.cmu.edu" -> null
    }

  # googleplay_user.siddharthr will be created
+ resource "googleplay_user" "siddharthr" {
+       email                = "sradhak2@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

Plan: 2 to add, 0 to change, 3 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 2 to add, 0 to change, 3 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.30s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.ops_siddharthr will be created
+ resource "google_cloud_identity_group_membership" "ops_siddharthr" {
+       create_time = (known after apply)
+       group       = "groups/03fwokq036zsp1f"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "sradhak2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

  # google_cloud_identity_group_membership.tech_acchs will be destroyed
  # (because google_cloud_identity_group_membership.tech_acchs is not in configuration)
- resource "google_cloud_identity_group_membership" "tech_acchs" {
-       create_time = "2026-09-06T04:03:48.051444Z" -> null
-       group       = "groups/04anzqyu1bqbv79" -> null
-       id          = "groups/04anzqyu1bqbv79/memberships/102039638856706939215" -> null
-       name        = "groups/04anzqyu1bqbv79/memberships/102039638856706939215" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-06T04:03:48.051444Z" -> null

-       preferred_member_key {
-           id = "azzhou@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.tech_siddharthr will be created
+ resource "google_cloud_identity_group_membership" "tech_siddharthr" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "sradhak2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 2 to add, 0 to change, 1 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 2 to add, 0 to change, 1 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.26s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-maps_maps will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-maps_maps" {
        id       = "scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64"
!       members  = [
-           "krishsax",
-           "spola2",
+           "sradhak2",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-maps_maps_admins will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-maps_maps_admins" {
        id       = "scottylabs/group-memberships/75db0c7f-7cfb-4d89-b84f-0f50523ed55b"
!       members  = [
-           "krishsax",
-           "spola2",
+           "sradhak2",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_occ_occ-main will be updated in-place
! resource "keycloak_group_memberships" "project_occ_occ-main" {
        id       = "scottylabs/group-memberships/a4e98eec-77cf-49f8-8ea6-32cf6332cf36"
!       members  = [
-           "azzhou",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
-           "azzhou",
+           "sradhak2",
            # (28 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 4 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 4 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.37s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)
- destroy

OpenTofu will perform the following actions:

  # discord_member_roles.acchs will be destroyed
  # (because discord_member_roles.acchs is not in configuration)
- resource "discord_member_roles" "acchs" {
-       id        = "1353118209024331948:781930226778767380" -> null
-       server_id = "1353118209024331948" -> null
-       user_id   = "781930226778767380" -> null

-       role {
-           has_role = true -> null
-           role_id  = "1488787088118644848" -> null
        }
-       role {
-           has_role = true -> null
-           role_id  = "1543486517908082728" -> null
        }
    }

  # discord_member_roles.krishsax will be updated in-place
! resource "discord_member_roles" "krishsax" {
        id        = "1353118209024331948:782966391321395240"
        # (2 unchanged attributes hidden)

-       role {
-           has_role = true -> null
-           role_id  = "1526003694708195448" -> null
        }

        # (3 unchanged blocks hidden)
    }

  # discord_member_roles.siddharthr will be created
+ resource "discord_member_roles" "siddharthr" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "584841698186756167"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1519792320755662928"
        }
+       role {
+           has_role = true
+           role_id  = "1526003694708195448"
        }
    }

  # discord_member_roles.xboxbedrock will be updated in-place
! resource "discord_member_roles" "xboxbedrock" {
        id        = "1353118209024331948:712398210345205871"
        # (2 unchanged attributes hidden)

-       role {
-           has_role = true -> null
-           role_id  = "1526003694708195448" -> null
        }

        # (5 unchanged blocks hidden)
    }

Plan: 1 to add, 2 to change, 1 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 2 to change, 1 to destroy.


6. project: git dir: tofu/git workspace: default

Plan Error

Show Output
running '/nix/store/v8n4x020rdfr57a0s3xdq46a3mdd5bk7-opentofu-1.12.5/bin/tofu plan -input=false -refresh -out /var/lib/atlantis/repos/ScottyLabs/governance/188/default/tofu/git/git-default.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/188/default/tofu/git': exit status 1
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.31s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
data.external.identity_jessica_lr_song: Reading...
data.external.identity_rzhu3: Reading...
data.external.identity_siddharthr: Reading...
data.external.identity_gostmeaper: Reading...
data.external.identity_liamwalk: Reading...
data.external.identity_jc_965: Reading...
data.external.identity_anish: Reading...
data.external.identity_siddharthr: Read complete after 1s [id=-]
data.external.identity_brightz: Reading...
data.external.identity_jc_965: Read complete after 1s [id=-]
data.external.identity_nikhilr: Reading...
data.external.identity_rzhu3: Read complete after 1s [id=-]
data.external.identity_liamwalk: Read complete after 1s [id=-]
data.external.identity_gostmeaper: Read complete after 1s [id=-]
data.external.identity_kothurvidur: Reading...
data.external.identity_jessica_lr_song: Read complete after 1s [id=-]
data.external.identity_jw8: Reading...
data.external.identity_amzoeee: Reading...
data.external.identity_dcrodrig: Reading...
data.external.identity_anish: Read complete after 1s [id=-]
data.external.identity_kechezon: Reading...
data.external.identity_maybe_yiyi: Reading...
data.external.identity_brightz: Read complete after 0s [id=-]
data.external.identity_dkou: Reading...
data.external.identity_zoexue: Reading...
data.external.identity_jw8: Read complete after 0s [id=-]
data.external.identity_kritdass: Reading...
data.external.identity_dcrodrig: Read complete after 0s [id=-]
data.external.identity_xboxbedrock: Reading...
data.external.identity_nikhilr: Read complete after 0s [id=-]
data.external.identity_jiyas: Reading...
data.external.identity_amzoeee: Read complete after 0s [id=-]
data.external.identity_manrajs: Reading...
data.external.identity_jalenluorion: Reading...
data.external.identity_kechezon: Read complete after 0s [id=-]
data.external.identity_tentype: Reading...
data.external.identity_maybe_yiyi: Read complete after 0s [id=-]
data.external.identity_thesuperrl: Reading...
data.external.identity_kothurvidur: Read complete after 0s [id=-]
data.external.identity_emmawong: Reading...
data.external.identity_dkou: Read complete after 0s [id=-]
data.external.identity_rhsen: Reading...
data.external.identity_zoexue: Read complete after 0s [id=-]
data.external.identity_etashjha: Reading...
data.external.identity_kritdass: Read complete after 0s [id=-]
data.external.identity_krishsax: Reading...
data.external.identity_jiyas: Read complete after 0s [id=-]
data.external.identity_riztofu: Reading...
data.external.identity_manrajs: Read complete after 0s [id=-]
random_password.kennel_webhook_secret: Refreshing state... [id=none]
github_repository.terrier_terrier: Refreshing state... [id=terrier]
data.external.identity_jalenluorion: Read complete after 0s [id=-]
github_team_membership.cmu-maps_xboxbedrock: Refreshing state... [id=10872360:xboxbedrock]
data.external.identity_xboxbedrock: Read complete after 0s [id=-]
github_team_membership.occ_acchs: Refreshing state... [id=19240310:acchs-at-cmu]
data.external.identity_tentype: Read complete after 0s [id=-]
github_team.cmu-courses: Refreshing state... [id=3643907]
data.external.identity_thesuperrl: Read complete after 0s [id=-]
github_repository.slai_cmugpt_agent: Refreshing state... [id=cmugpt-agent]
data.external.identity_emmawong: Read complete after 0s [id=-]
data.external.identity_rhsen: Read complete after 1s [id=-]
github_team.cmu-eats: Refreshing state... [id=16807600]
github_team.cbp: Refreshing state... [id=17170099]
github_repository.cbp_groupme_mirror: Refreshing state... [id=groupme-mirror]
data.external.identity_etashjha: Read complete after 1s [id=-]
github_repository.devops_collie: Refreshing state... [id=collie]
data.external.identity_krishsax: Read complete after 1s [id=-]
data.external.identity_riztofu: Read complete after 1s [id=-]
github_repository.cmu-housing_housing: Refreshing state... [id=housing]
github_team.uia: Refreshing state... [id=17529763]
github_team.cmu-housing: Refreshing state... [id=17937919]
github_team.occ: Refreshing state... [id=19240310]
github_team.cmu-maps: Refreshing state... [id=10872360]
github_repository.devops_infrastructure: Refreshing state... [id=infrastructure]
github_team.cmu-cal: Refreshing state... [id=15669112]
github_team.tartan-vote: Refreshing state... [id=17701833]
github_team.slai: Refreshing state... [id=17571328]
github_repository.cmu-insta_insta: Refreshing state... [id=insta]
github_repository.cmu-courses_internet_archive: Refreshing state... [id=internet-archive]
github_repository.tartan-vote_tartan_vote: Refreshing state... [id=tartan-vote]
github_team_membership.cmu-maps_krishsax: Refreshing state... [id=10872360:krishsaxena]
github_team.cmu-insta: Refreshing state... [id=19240311]
github_repository.devops_governance: Refreshing state... [id=governance]
github_repository.slai_cmugpt_surface: Refreshing state... [id=cmugpt-surface]
github_repository.devops_link_shortener: Refreshing state... [id=link-shortener]
github_repository.cmu-eats_eats: Refreshing state... [id=eats]
github_team.lost-and-found: Refreshing state... [id=4088608]
github_repository.slai_cmugpt_sms_surface: Refreshing state... [id=cmugpt-sms-surface]
github_repository.slai_mcp_server: Refreshing state... [id=mcp-server]
github_repository.uia_scottylabs_org: Refreshing state... [id=scottylabs-org]
github_repository.cmu-eats_dining_api: Refreshing state... [id=dining-api]
github_team.terrier: Refreshing state... [id=12028577]
github_repository.cmu-courses_courses: Refreshing state... [id=courses]
github_repository.devops_kennel: Refreshing state... [id=kennel]
github_repository.uia_components: Refreshing state... [id=components]
github_repository.cmu-cal_cal: Refreshing state... [id=cal]
github_repository.cbp_bus_sign: Refreshing state... [id=bus-sign]
github_repository.devops_documentation: Refreshing state... [id=documentation]
github_repository.quest_quest: Refreshing state... [id=quest]
github_repository.lost-and-found_lost_and_found_v2: Refreshing state... [id=lost-and-found-v2]
github_repository.occ_occ: Refreshing state... [id=occ]
github_repository.cbp_discord_verify: Refreshing state... [id=discord-verify]
github_team.quest: Refreshing state... [id=17330197]
github_team.devops: Refreshing state... [id=14706218]
github_repository.cbp_dalmatian: Refreshing state... [id=dalmatian]
github_repository.cmu-maps_maps: Refreshing state... [id=maps]
forgejo_repository.cmu-housing_housing: Refreshing state... [name=housing]
forgejo_team_member.occ_acchs: Refreshing state...
forgejo_repository.devops_infrastructure: Refreshing state... [name=infrastructure]
forgejo_team_member.cmu-maps_krishsax: Refreshing state...
forgejo_team_member.cmu-maps_xboxbedrock: Refreshing state...
forgejo_repository.cmu-maps_maps: Refreshing state... [name=maps]
forgejo_repository.cmu-eats_eats: Refreshing state... [name=eats]
forgejo_repository.occ_occ: Refreshing state... [name=occ]
forgejo_repository.cmu-insta_insta: Refreshing state... [name=insta]
forgejo_repository.devops_kennel: Refreshing state... [name=kennel]
forgejo_repository.cbp_groupme_mirror: Refreshing state... [name=groupme-mirror]
forgejo_repository.lost-and-found_lost_and_found_v2: Refreshing state... [name=lost-and-found-v2]
forgejo_repository.slai_mcp_server: Refreshing state... [name=mcp-server]
forgejo_team_member.maps_leads_krishsax: Refreshing state...
forgejo_repository.slai_cmugpt_sms_surface: Refreshing state... [name=cmugpt-sms-surface]
forgejo_repository.tartan-vote_tartan_vote: Refreshing state... [name=tartan-vote]
forgejo_repository.devops_documentation: Refreshing state... [name=documentation]
forgejo_team_member.maps_leads_xboxbedrock: Refreshing state...
forgejo_repository.uia_components: Refreshing state... [name=components]
forgejo_repository.cmu-cal_cal: Refreshing state... [name=cal]
forgejo_repository.cbp_dalmatian: Refreshing state... [name=dalmatian]
forgejo_repository.quest_quest: Refreshing state... [name=quest]
forgejo_repository.cbp_discord_verify: Refreshing state... [name=discord-verify]
forgejo_repository.cmu-courses_internet_archive: Refreshing state... [name=internet-archive]
data.forgejo_organization.this: Reading...
forgejo_repository.devops_link_shortener: Refreshing state... [name=link-shortener]
data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs]
forgejo_repository.devops_collie: Refreshing state... [name=collie]
forgejo_repository.devops_governance: Refreshing state... [name=governance]
forgejo_repository.cmu-courses_courses: Refreshing state... [name=courses]
forgejo_repository.cbp_bus_sign: Refreshing state... [name=bus-sign]
forgejo_repository.terrier_terrier: Refreshing state... [name=terrier]
forgejo_repository.slai_cmugpt_surface: Refreshing state... [name=cmugpt-surface]
forgejo_repository.uia_scottylabs_org: Refreshing state... [name=scottylabs-org]
forgejo_repository.slai_cmugpt_agent: Refreshing state... [name=cmugpt-agent]
forgejo_repository.cmu-eats_dining_api: Refreshing state... [name=dining-api]
github_actions_repository_permissions.terrier_terrier_actions: Refreshing state... [id=terrier]
github_team_membership.cmu-courses_anish: Refreshing state... [id=3643907:ap-1]
github_team_membership.cmu-courses_etashjha: Refreshing state... [id=3643907:etashj]
github_actions_repository_permissions.slai_cmugpt_agent_actions: Refreshing state... [id=cmugpt-agent]
github_team_membership.cmu-eats_anish: Refreshing state... [id=16807600:ap-1]
github_team_membership.cbp_liamwalk: Refreshing state... [id=17170099:pandaaured]
github_team_membership.cbp_maybe_yiyi: Refreshing state... [id=17170099:maybe-yiyi]
github_team_membership.cbp_tentype: Refreshing state... [id=17170099:tentype]
github_actions_repository_permissions.cbp_groupme_mirror_actions: Refreshing state... [id=groupme-mirror]
github_actions_repository_permissions.devops_collie_actions: Refreshing state... [id=collie]
github_team_membership.cmu-housing_emmawong: Refreshing state... [id=17937919:emw868]
github_team_membership.cmu-housing_rhsen: Refreshing state... [id=17937919:rohansen08]
github_team_membership.cmu-housing_gostmeaper: Refreshing state... [id=17937919:gostmeaper]
github_team_membership.cmu-housing_manrajs: Refreshing state... [id=17937919:manraj10]
github_team_membership.cmu-housing_nikhilr: Refreshing state... [id=17937919:ecstaticpilot]
github_team_membership.cmu-housing_tentype: Refreshing state... [id=17937919:tentype]
github_team_membership.cmu-housing_zoexue: Refreshing state... [id=17937919:o-zox-o]
github_team_membership.cmu-housing_xboxbedrock: Refreshing state... [id=17937919:xboxbedrock]
github_team_membership.cmu-housing_amzoeee: Refreshing state... [id=17937919:amzoeee]
github_team_membership.cmu-housing_jessica_lr_song: Refreshing state... [id=17937919:jessica-lr-song]
github_team_membership.occ_anish: Refreshing state... [id=19240310:ap-1]
github_team_membership.occ_dcrodrig: Refreshing state... [id=19240310:newhacker1746]
github_team_membership.cmu-maps_anish: Refreshing state... [id=10872360:ap-1]
github_team_membership.cmu-cal_jalenluorion: Refreshing state... [id=15669112:jalenluorion]
github_team_membership.tartan-vote_maybe_yiyi: Refreshing state... [id=17701833:maybe-yiyi]
github_team_membership.tartan-vote_riztofu: Refreshing state... [id=17701833:riztofu]
github_team_membership.slai_rzhu3: Refreshing state... [id=17571328:richard-31415]
github_team_membership.slai_krishsax: Refreshing state... [id=17571328:krishsaxena]
github_team_membership.slai_jc_965: Refreshing state... [id=17571328:jc-965]
github_team_membership.slai_jiyas: Refreshing state... [id=17571328:jiya2301]
github_actions_repository_permissions.cmu-courses_internet_archive_actions: Refreshing state... [id=internet-archive]
github_actions_repository_permissions.cmu-housing_housing_actions: Refreshing state... [id=housing]
github_team_membership.cmu-insta_etashjha: Refreshing state... [id=19240311:etashj]
github_actions_repository_permissions.devops_infrastructure_actions: Refreshing state... [id=infrastructure]
github_actions_repository_permissions.slai_cmugpt_surface_actions: Refreshing state... [id=cmugpt-surface]
github_actions_repository_permissions.cmu-insta_insta_actions: Refreshing state... [id=insta]
github_actions_repository_permissions.tartan-vote_tartan_vote_actions: Refreshing state... [id=tartan-vote]
github_actions_repository_permissions.slai_cmugpt_sms_surface_actions: Refreshing state... [id=cmugpt-sms-surface]
github_actions_repository_permissions.devops_governance_actions: Refreshing state... [id=governance]
github_actions_repository_permissions.devops_link_shortener_actions: Refreshing state... [id=link-shortener]
github_team_membership.terrier_dkou: Refreshing state... [id=12028577:sharkree]
github_team_membership.terrier_kritdass: Refreshing state... [id=12028577:kritdass]
github_team_membership.terrier_etashjha: Refreshing state... [id=12028577:etashj]
github_team_membership.terrier_kothurvidur: Refreshing state... [id=12028577:kothurvidur]
github_team_membership.terrier_anish: Refreshing state... [id=12028577:ap-1]
github_actions_repository_permissions.cmu-eats_eats_actions: Refreshing state... [id=eats]
github_actions_repository_permissions.cmu-courses_courses_actions: Refreshing state... [id=courses]
github_actions_repository_permissions.uia_components_actions: Refreshing state... [id=components]
github_actions_repository_permissions.slai_mcp_server_actions: Refreshing state... [id=mcp-server]
github_actions_repository_permissions.uia_scottylabs_org_actions: Refreshing state... [id=scottylabs-org]
github_actions_repository_permissions.cmu-eats_dining_api_actions: Refreshing state... [id=dining-api]
github_actions_repository_permissions.devops_documentation_actions: Refreshing state... [id=documentation]
github_actions_repository_permissions.lost-and-found_lost_and_found_v2_actions: Refreshing state... [id=lost-and-found-v2]
github_actions_repository_permissions.devops_kennel_actions: Refreshing state... [id=kennel]
restapi_object.cmu-housing_housing_push_mirror: Refreshing state... [id=remote_mirror_uoV2HG4aiOT]
restapi_object.cmu-housing_housing_docs_webhook: Refreshing state... [id=15]
restapi_object.devops_infrastructure_push_mirror: Refreshing state... [id=remote_mirror_oYtFpKVzTaZ]
restapi_object.cmu-maps_maps_push_mirror: Refreshing state... [id=remote_mirror_e4_FUnvoslJ]
restapi_object.cmu-eats_eats_push_mirror: Refreshing state... [id=remote_mirror_sjYwMA6NPgh]
restapi_object.occ_occ_push_mirror: Refreshing state... [id=remote_mirror_moBLq1YG1qz]
restapi_object.cmu-insta_insta_push_mirror: Refreshing state... [id=remote_mirror_891nu6XQ1us]
restapi_object.devops_kennel_push_mirror: Refreshing state... [id=remote_mirror_cT9psJkhTQ4]
restapi_object.cbp_groupme_mirror_push_mirror: Refreshing state... [id=remote_mirror_U61HJHB9VGE]
restapi_object.lost-and-found_lost_and_found_v2_push_mirror: Refreshing state... [id=remote_mirror_n_UZBFU-CY4]
restapi_object.slai_mcp_server_push_mirror: Refreshing state... [id=remote_mirror_Tqh2y0R_vDp]
restapi_object.slai_cmugpt_sms_surface_push_mirror: Refreshing state... [id=remote_mirror__fkBkASN2jA]
restapi_object.tartan-vote_tartan_vote_push_mirror: Refreshing state... [id=remote_mirror_EUFq4sFSnH5]
restapi_object.tartan-vote_tartan_vote_docs_webhook: Refreshing state... [id=25]
restapi_object.devops_documentation_push_mirror: Refreshing state... [id=remote_mirror_izmsqaNB6Wx]
github_actions_repository_permissions.cbp_discord_verify_actions: Refreshing state... [id=discord-verify]
restapi_object.uia_components_push_mirror: Refreshing state... [id=remote_mirror_Kwwa8TDnTgc]
restapi_object.cmu-cal_cal_push_mirror: Refreshing state... [id=remote_mirror_CO243uIONoq]
restapi_object.cbp_dalmatian_push_mirror: Refreshing state... [id=remote_mirror_qzlCpLK153F]
restapi_object.cbp_dalmatian_docs_webhook: Refreshing state... [id=7]
restapi_object.quest_quest_push_mirror: Refreshing state... [id=remote_mirror_U9d9BPAon29]
restapi_object.cmu-courses_internet_archive_push_mirror: Refreshing state... [id=remote_mirror_58X__Yfs0bR]
restapi_object.cbp_discord_verify_push_mirror: Refreshing state... [id=remote_mirror_3MV7Up6cY1X]
github_team_membership.quest_anish: Refreshing state... [id=17330197:ap-1]
github_team_membership.quest_kechezon: Refreshing state... [id=17330197:kechezon]
github_team_membership.quest_brightz: Refreshing state... [id=17330197:bright-zheng]
github_team_membership.quest_xboxbedrock: Refreshing state... [id=17330197:xboxbedrock]
github_team_membership.quest_jw8: Refreshing state... [id=17330197:jeffreyyijiwang]
forgejo_team.terrier: Refreshing state... [name=terrier]
forgejo_team.lost-and-found: Refreshing state... [name=lost-and-found]
forgejo_team.tartan-vote_leads: Refreshing state... [name=tartan-vote-leads]
forgejo_team.dalmatian_leads: Refreshing state... [name=dalmatian-leads]
forgejo_team.insta_leads: Refreshing state... [name=insta-leads]
forgejo_team.cmu-insta: Refreshing state... [name=cmu-insta]
forgejo_team.cmu-maps: Refreshing state... [name=cmu-maps]
forgejo_team.devops: Refreshing state... [name=devops]
forgejo_team.maps_leads: Refreshing state... [name=maps-leads]
forgejo_team.eats_leads: Refreshing state... [name=eats-leads]
forgejo_team.tartan-vote: Refreshing state... [name=tartan-vote]
forgejo_team.groupme-mirror_leads: Refreshing state... [name=groupme-mirror-leads]
forgejo_team.cmu-eats: Refreshing state... [name=cmu-eats]
forgejo_team.bus-sign_leads: Refreshing state... [name=bus-sign-leads]
forgejo_team.cbp: Refreshing state... [name=cbp]
forgejo_team.terrier_leads: Refreshing state... [name=terrier-leads]
forgejo_team.discord-verify_leads: Refreshing state... [name=discord-verify-leads]
forgejo_team.quest: Refreshing state... [name=quest]
forgejo_team.occ: Refreshing state... [name=occ]
forgejo_team.cmugpt-surface_leads: Refreshing state... [name=cmugpt-surface-leads]
forgejo_team.documentation_leads: Refreshing state... [name=documentation-leads]
forgejo_team.courses_leads: Refreshing state... [name=courses-leads]
forgejo_team.devops_leads: Refreshing state... [name=devops-leads]
forgejo_team.cmugpt-mcp-server_leads: Refreshing state... [name=cmugpt-mcp-server-leads]
forgejo_team.link-shortener_leads: Refreshing state... [name=link-shortener-leads]
forgejo_team.housing_leads: Refreshing state... [name=housing-leads]
forgejo_team.occ-main_leads: Refreshing state... [name=occ-main-leads]
forgejo_team.slai: Refreshing state... [name=slai]
forgejo_team.cmu-courses: Refreshing state... [name=cmu-courses]
forgejo_team.cmu-courses_leads: Refreshing state... [name=cmu-courses-leads]
forgejo_team.cal_leads: Refreshing state... [name=cal-leads]
forgejo_team.cmugpt-sms-surface_leads: Refreshing state... [name=cmugpt-sms-surface-leads]
forgejo_team.uia: Refreshing state... [name=uia]
forgejo_team.quest_leads: Refreshing state... [name=quest-leads]
forgejo_team.cmugpt-agent_leads: Refreshing state... [name=cmugpt-agent-leads]
forgejo_team.cmu-housing: Refreshing state... [name=cmu-housing]
forgejo_team.cmu-cal: Refreshing state... [name=cmu-cal]
restapi_object.devops_link_shortener_push_mirror: Refreshing state... [id=remote_mirror_cp8daMpt82v]
restapi_object.devops_collie_push_mirror: Refreshing state... [id=remote_mirror_ljvVejNFXaS]
restapi_object.devops_governance_push_mirror: Refreshing state... [id=remote_mirror_lt9JzkX1jMq]
restapi_object.cmu-courses_courses_push_mirror: Refreshing state... [id=remote_mirror_WMR9yQwKFDF]
restapi_object.cbp_bus_sign_push_mirror: Refreshing state... [id=remote_mirror_MkKxQ2vA9pc]
restapi_object.terrier_terrier_push_mirror: Refreshing state... [id=remote_mirror_YopDgZksZv5]
restapi_object.slai_cmugpt_surface_push_mirror: Refreshing state... [id=remote_mirror_cPTQN6S5SpB]
github_actions_repository_permissions.cmu-cal_cal_actions: Refreshing state... [id=cal]
restapi_object.cmu-courses_courses_kennel_webhook: Refreshing state... [id=11]
restapi_object.cmu-eats_eats_kennel_webhook: Refreshing state... [id=31]
restapi_object.cbp_discord_verify_kennel_webhook: Refreshing state... [id=18]
restapi_object.uia_components_kennel_webhook: Refreshing state... [id=27]
restapi_object.slai_cmugpt_surface_kennel_webhook: Refreshing state... [id=23]
restapi_object.cmu-housing_housing_kennel_webhook: Refreshing state... [id=16]
restapi_object.devops_kennel_kennel_webhook: Refreshing state... [id=14]
restapi_object.terrier_terrier_kennel_webhook: Refreshing state... [id=12]
restapi_object.cmu-maps_maps_kennel_webhook: Refreshing state... [id=9]
restapi_object.cmu-cal_cal_kennel_webhook: Refreshing state... [id=20]
restapi_object.cbp_bus_sign_kennel_webhook: Refreshing state... [id=21]
restapi_object.tartan-vote_tartan_vote_kennel_webhook: Refreshing state... [id=24]
restapi_object.devops_link_shortener_kennel_webhook: Refreshing state... [id=8]
restapi_object.devops_infrastructure_kennel_webhook: Refreshing state... [id=26]
restapi_object.devops_governance_kennel_webhook: Refreshing state... [id=22]
restapi_object.slai_cmugpt_sms_surface_kennel_webhook: Refreshing state... [id=19]
restapi_object.quest_quest_kennel_webhook: Refreshing state... [id=10]
restapi_object.slai_cmugpt_agent_push_mirror: Refreshing state... [id=remote_mirror_kNY6rVcxdyr]
restapi_object.slai_cmugpt_agent_kennel_webhook: Refreshing state... [id=28]
restapi_object.slai_mcp_server_kennel_webhook: Refreshing state... [id=13]
restapi_object.devops_collie_kennel_webhook: Refreshing state... [id=17]
restapi_object.cbp_dalmatian_kennel_webhook: Refreshing state... [id=6]
restapi_object.cmu-eats_dining_api_push_mirror: Refreshing state... [id=remote_mirror_a4MsQ5rO1_Z]
restapi_object.uia_scottylabs_org_push_mirror: Refreshing state... [id=remote_mirror_h6yN6QdS1lD]
restapi_object.uia_scottylabs_org_kennel_webhook: Refreshing state... [id=29]
github_team_membership.devops_anish: Refreshing state... [id=14706218:ap-1]
github_team_membership.devops_xboxbedrock: Refreshing state... [id=14706218:xboxbedrock]
github_team_membership.devops_thesuperrl: Refreshing state... [id=14706218:thesuperrl]
github_actions_repository_permissions.cbp_bus_sign_actions: Refreshing state... [id=bus-sign]
github_actions_repository_permissions.quest_quest_actions: Refreshing state... [id=quest]
github_actions_repository_permissions.occ_occ_actions: Refreshing state... [id=occ]
github_actions_repository_permissions.cbp_dalmatian_actions: Refreshing state... [id=dalmatian]
github_actions_repository_permissions.cmu-maps_maps_actions: Refreshing state... [id=maps]
github_repository_deploy_key.cmu-housing_housing_mirror_deploy_key: Refreshing state... [id=housing:159665369]
github_repository_deploy_key.devops_infrastructure_mirror_deploy_key: Refreshing state... [id=infrastructure:159665343]
github_repository_deploy_key.cmu-maps_maps_mirror_deploy_key: Refreshing state... [id=maps:159665325]
github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Refreshing state... [id=eats:161775995]
github_repository_deploy_key.occ_occ_mirror_deploy_key: Refreshing state... [id=occ:161720242]
github_repository_deploy_key.cmu-insta_insta_mirror_deploy_key: Refreshing state... [id=insta:161720245]
github_repository_deploy_key.devops_kennel_mirror_deploy_key: Refreshing state... [id=kennel:159665328]
github_repository_deploy_key.cbp_groupme_mirror_mirror_deploy_key: Refreshing state... [id=groupme-mirror:159665366]
github_repository_deploy_key.lost-and-found_lost_and_found_v2_mirror_deploy_key: Refreshing state... [id=lost-and-found-v2:159665337]
github_repository_deploy_key.slai_mcp_server_mirror_deploy_key: Refreshing state... [id=mcp-server:159665358]
github_repository_deploy_key.slai_cmugpt_sms_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-sms-surface:159665365]
github_repository_deploy_key.tartan-vote_tartan_vote_mirror_deploy_key: Refreshing state... [id=tartan-vote:159665350]
github_repository_deploy_key.devops_documentation_mirror_deploy_key: Refreshing state... [id=documentation:159665362]
github_repository_deploy_key.uia_components_mirror_deploy_key: Refreshing state... [id=components:159665353]
github_repository_deploy_key.cbp_dalmatian_mirror_deploy_key: Refreshing state... [id=dalmatian:159665367]
github_repository_deploy_key.cmu-cal_cal_mirror_deploy_key: Refreshing state... [id=cal:159665344]
github_repository_deploy_key.quest_quest_mirror_deploy_key: Refreshing state... [id=quest:159665323]
github_repository_deploy_key.cmu-courses_internet_archive_mirror_deploy_key: Refreshing state... [id=internet-archive:159665356]
github_repository_deploy_key.cbp_discord_verify_mirror_deploy_key: Refreshing state... [id=discord-verify:159665340]
forgejo_team_member.terrier_kothurvidur: Refreshing state...
forgejo_team_member.terrier_anish: Refreshing state...
forgejo_team_member.terrier_kritdass: Refreshing state...
forgejo_team_repository.terrier_terrier: Refreshing state...
forgejo_team_member.terrier_etashjha: Refreshing state...
forgejo_team_member.terrier_dkou: Refreshing state...
forgejo_team_repository.lost-and-found_lost_and_found_v2: Refreshing state...
forgejo_team_member.tartan-vote_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.dalmatian_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.dalmatian_leads_tentype: Refreshing state...
forgejo_team_member.insta_leads_etashjha: Refreshing state...
forgejo_team_member.cmu-insta_etashjha: Refreshing state...
forgejo_team_repository.cmu-insta_insta: Refreshing state...
forgejo_team_repository.cmu-maps_maps: Refreshing state...
forgejo_team_member.cmu-maps_anish: Refreshing state...
forgejo_team_repository.devops_governance: Refreshing state...
forgejo_team_repository.devops_collie: Refreshing state...
forgejo_team_repository.devops_link_shortener: Refreshing state...
forgejo_team_member.devops_xboxbedrock: Refreshing state...
forgejo_team_repository.devops_documentation: Refreshing state...
forgejo_team_member.devops_thesuperrl: Refreshing state...
forgejo_team_repository.devops_infrastructure: Refreshing state...
forgejo_team_member.devops_anish: Refreshing state...
forgejo_team_repository.devops_kennel: Refreshing state...
forgejo_team_member.maps_leads_anish: Refreshing state...
forgejo_team_member.tartan-vote_riztofu: Refreshing state...
forgejo_team_member.eats_leads_anish: Refreshing state...
forgejo_team_member.tartan-vote_maybe_yiyi: Refreshing state...
forgejo_team_repository.tartan-vote_tartan_vote: Refreshing state...
forgejo_team_member.groupme-mirror_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.groupme-mirror_leads_tentype: Refreshing state...
forgejo_team_repository.cmu-eats_eats: Refreshing state...
forgejo_team_member.cmu-eats_anish: Refreshing state...
forgejo_team_repository.cmu-eats_dining_api: Refreshing state...
forgejo_team_member.bus-sign_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.bus-sign_leads_tentype: Refreshing state...
forgejo_team_member.cbp_tentype: Refreshing state...
forgejo_team_repository.cbp_discord_verify: Refreshing state...
forgejo_team_repository.cbp_dalmatian: Refreshing state...
forgejo_team_repository.cbp_groupme_mirror: Refreshing state...
forgejo_team_member.cbp_maybe_yiyi: Refreshing state...
forgejo_team_repository.cbp_bus_sign: Refreshing state...
forgejo_team_member.cbp_liamwalk: Refreshing state...
forgejo_team_member.terrier_leads_kritdass: Refreshing state...
forgejo_team_member.terrier_leads_anish: Refreshing state...
forgejo_team_member.discord-verify_leads_tentype: Refreshing state...
forgejo_team_member.discord-verify_leads_maybe_yiyi: Refreshing state...
forgejo_team_repository.quest_quest: Refreshing state...
forgejo_team_member.quest_brightz: Refreshing state...
forgejo_team_member.quest_jw8: Refreshing state...
forgejo_team_member.quest_kechezon: Refreshing state...
forgejo_team_member.quest_anish: Refreshing state...
forgejo_team_member.quest_xboxbedrock: Refreshing state...
forgejo_team_member.occ_anish: Refreshing state...
forgejo_team_repository.occ_occ: Refreshing state...
forgejo_team_member.occ_dcrodrig: Refreshing state...
forgejo_team_member.cmugpt-surface_leads_krishsax: Refreshing state...
forgejo_team_member.documentation_leads_thesuperrl: Refreshing state...
forgejo_team_member.documentation_leads_anish: Refreshing state...
forgejo_team_member.devops_leads_anish: Refreshing state...
forgejo_team_member.devops_leads_thesuperrl: Refreshing state...
forgejo_team_member.courses_leads_anish: Refreshing state...
forgejo_team_member.courses_leads_etashjha: Refreshing state...
forgejo_team_member.cmugpt-mcp-server_leads_krishsax: Refreshing state...
forgejo_team_member.link-shortener_leads_anish: Refreshing state...
forgejo_team_member.link-shortener_leads_thesuperrl: Refreshing state...
forgejo_team_member.housing_leads_tentype: Refreshing state...
forgejo_team_member.housing_leads_emmawong: Refreshing state...
forgejo_team_member.housing_leads_nikhilr: Refreshing state...
forgejo_team_member.housing_leads_gostmeaper: Refreshing state...
forgejo_team_member.occ-main_leads_dcrodrig: Refreshing state...
forgejo_team_member.occ-main_leads_anish: Refreshing state...
forgejo_team_member.slai_jiyas: Refreshing state...
forgejo_team_repository.slai_mcp_server: Refreshing state...
forgejo_team_repository.slai_cmugpt_agent: Refreshing state...
forgejo_team_repository.slai_cmugpt_surface: Refreshing state...
forgejo_team_member.slai_krishsax: Refreshing state...
forgejo_team_member.slai_jc_965: Refreshing state...
forgejo_team_repository.slai_cmugpt_sms_surface: Refreshing state...
forgejo_team_member.slai_rzhu3: Refreshing state...
forgejo_team_member.cmu-courses_leads_etashjha: Refreshing state...
forgejo_team_member.cmu-courses_leads_anish: Refreshing state...
forgejo_team_member.cmu-courses_anish: Refreshing state...
forgejo_team_repository.cmu-courses_internet_archive: Refreshing state...
forgejo_team_repository.cmu-courses_courses: Refreshing state...
forgejo_team_member.cmu-courses_etashjha: Refreshing state...
forgejo_team_member.cal_leads_jalenluorion: Refreshing state...
forgejo_team_member.cmugpt-sms-surface_leads_krishsax: Refreshing state...
forgejo_team_member.quest_leads_anish: Refreshing state...
forgejo_team_member.quest_leads_xboxbedrock: Refreshing state...
forgejo_team_repository.uia_scottylabs_org: Refreshing state...
forgejo_team_repository.cmu-housing_housing: Refreshing state...
forgejo_team_repository.uia_components: Refreshing state...
forgejo_team_member.cmu-housing_tentype: Refreshing state...
forgejo_team_member.cmu-housing_emmawong: Refreshing state...
forgejo_team_member.cmu-housing_rhsen: Refreshing state...
forgejo_team_member.cmu-housing_xboxbedrock: Refreshing state...
forgejo_team_member.cmu-housing_manrajs: Refreshing state...
forgejo_team_member.cmu-housing_zoexue: Refreshing state...
forgejo_team_member.cmu-housing_nikhilr: Refreshing state...
forgejo_team_member.cmu-housing_jessica_lr_song: Refreshing state...
forgejo_team_member.cmu-housing_gostmeaper: Refreshing state...
forgejo_team_member.cmu-housing_amzoeee: Refreshing state...
forgejo_team_repository.cmu-cal_cal: Refreshing state...
forgejo_team_member.cmu-cal_jalenluorion: Refreshing state...
forgejo_team_member.cmugpt-agent_leads_krishsax: Refreshing state...
github_repository_deploy_key.devops_link_shortener_mirror_deploy_key: Refreshing state... [id=link-shortener:159665345]
github_repository_deploy_key.devops_collie_mirror_deploy_key: Refreshing state... [id=collie:159665332]
github_repository_deploy_key.devops_governance_mirror_deploy_key: Refreshing state... [id=governance:159665346]
github_repository_deploy_key.cmu-courses_courses_mirror_deploy_key: Refreshing state... [id=courses:159665327]
github_repository_deploy_key.terrier_terrier_mirror_deploy_key: Refreshing state... [id=terrier:159665339]
github_repository_deploy_key.cbp_bus_sign_mirror_deploy_key: Refreshing state... [id=bus-sign:159665342]
github_repository_deploy_key.slai_cmugpt_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-surface:159665348]
github_repository_deploy_key.slai_cmugpt_agent_mirror_deploy_key: Refreshing state... [id=cmugpt-agent:159665360]
github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Refreshing state... [id=dining-api:161772815]
github_repository_deploy_key.uia_scottylabs_org_mirror_deploy_key: Refreshing state... [id=scottylabs-org:159665361]
github_repository_ruleset.cmu-housing_housing_ruleset: Refreshing state... [id=17421604]
github_repository_ruleset.devops_infrastructure_ruleset: Refreshing state... [id=11275118]
github_repository_ruleset.cmu-maps_maps_ruleset: Refreshing state... [id=18864961]
github_repository_ruleset.cmu-eats_eats_ruleset: Refreshing state... [id=21330218]
github_repository_ruleset.occ_occ_ruleset: Refreshing state... [id=21841634]
github_repository_ruleset.cmu-insta_insta_ruleset: Refreshing state... [id=21841636]
github_repository_ruleset.devops_kennel_ruleset: Refreshing state... [id=13193453]
github_repository_ruleset.cbp_groupme_mirror_ruleset: Refreshing state... [id=12503296]
github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Refreshing state... [id=18920618]
github_repository_ruleset.slai_mcp_server_ruleset: Refreshing state... [id=17082154]
forgejo_team_repository.terrier_leads_terrier: Refreshing state...
github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Refreshing state... [id=17653622]
forgejo_team_repository.insta_leads_insta: Refreshing state...
forgejo_team_repository.maps_leads_maps: Refreshing state...
forgejo_team_repository.devops_leads_governance: Refreshing state...
forgejo_team_repository.devops_leads_collie: Refreshing state...
forgejo_team_repository.link-shortener_leads_link_shortener: Refreshing state...
forgejo_team_repository.documentation_leads_documentation: Refreshing state...
forgejo_team_repository.devops_leads_infrastructure: Refreshing state...
github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Refreshing state... [id=16848966]
forgejo_team_repository.devops_leads_kennel: Refreshing state...
forgejo_team_repository.tartan-vote_leads_tartan_vote: Refreshing state...
forgejo_team_repository.eats_leads_eats: Refreshing state...
forgejo_team_repository.eats_leads_dining_api: Refreshing state...
forgejo_team_repository.discord-verify_leads_discord_verify: Refreshing state...
forgejo_team_repository.dalmatian_leads_dalmatian: Refreshing state...
forgejo_team_repository.groupme-mirror_leads_groupme_mirror: Refreshing state...
forgejo_team_repository.bus-sign_leads_bus_sign: Refreshing state...
github_repository_ruleset.devops_documentation_ruleset: Refreshing state... [id=17079368]
forgejo_team_repository.quest_leads_quest: Refreshing state...
forgejo_team_repository.occ-main_leads_occ: Refreshing state...
github_repository_ruleset.uia_components_ruleset: Refreshing state... [id=16366085]
forgejo_team_repository.cmugpt-agent_leads_cmugpt_agent: Refreshing state...
forgejo_team_repository.cmugpt-surface_leads_cmugpt_surface: Refreshing state...
forgejo_team_repository.cmugpt-mcp-server_leads_mcp_server: Refreshing state...
forgejo_team_repository.cmugpt-sms-surface_leads_cmugpt_sms_surface: Refreshing state...
forgejo_team_repository.cmu-courses_leads_internet_archive: Refreshing state...
forgejo_team_repository.courses_leads_courses: Refreshing state...
forgejo_team_repository.housing_leads_housing: Refreshing state...
forgejo_team_repository.cal_leads_cal: Refreshing state...
github_repository_ruleset.cbp_dalmatian_ruleset: Refreshing state... [id=17238915]
github_repository_ruleset.cmu-cal_cal_ruleset: Refreshing state... [id=18919281]
github_repository_ruleset.quest_quest_ruleset: Refreshing state... [id=15750052]
github_repository_ruleset.cmu-courses_internet_archive_ruleset: Refreshing state... [id=11988074]
github_repository_ruleset.cbp_discord_verify_ruleset: Refreshing state... [id=11449154]
github_repository_ruleset.devops_governance_ruleset: Refreshing state... [id=14488792]
github_repository_ruleset.devops_collie_ruleset: Refreshing state... [id=18865204]
github_repository_ruleset.devops_link_shortener_ruleset: Refreshing state... [id=17911647]
github_repository_ruleset.cmu-courses_courses_ruleset: Refreshing state... [id=15855161]
github_repository_ruleset.terrier_terrier_ruleset: Refreshing state... [id=12671003]
github_repository_ruleset.cbp_bus_sign_ruleset: Refreshing state... [id=15236550]
github_repository_ruleset.slai_cmugpt_surface_ruleset: Refreshing state... [id=16476662]
github_repository_ruleset.slai_cmugpt_agent_ruleset: Refreshing state... [id=16476661]
github_repository_ruleset.cmu-eats_dining_api_ruleset: Refreshing state... [id=21879626]
github_repository_ruleset.uia_scottylabs_org_ruleset: Refreshing state... [id=18112214]

OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
  + create
  ~ update in-place (current -> planned)
  - destroy

OpenTofu planned the following actions, but then encountered a problem:

  # forgejo_team_member.cmu-maps_siddharthr will be created
  + resource "forgejo_team_member" "cmu-maps_siddharthr" {
      + team_id = 13
      + user    = "sradhak2"
    }

  # forgejo_team_member.cmu-maps_xboxbedrock will be destroyed
  # (because forgejo_team_member.cmu-maps_xboxbedrock is not in configuration)
  - resource "forgejo_team_member" "cmu-maps_xboxbedrock" {
      - team_id = 13 -> null
      - user    = "xboxbedrock" -> null
    }

  # forgejo_team_member.maps_leads_krishsax will be destroyed
  # (because forgejo_team_member.maps_leads_krishsax is not in configuration)
  - resource "forgejo_team_member" "maps_leads_krishsax" {
      - team_id = 38 -> null
      - user    = "krishsax" -> null
    }

  # forgejo_team_member.maps_leads_siddharthr will be created
  + resource "forgejo_team_member" "maps_leads_siddharthr" {
      + team_id = 38
      + user    = "sradhak2"
    }

  # forgejo_team_member.maps_leads_xboxbedrock will be destroyed
  # (because forgejo_team_member.maps_leads_xboxbedrock is not in configuration)
  - resource "forgejo_team_member" "maps_leads_xboxbedrock" {
      - team_id = 38 -> null
      - user    = "xboxbedrock" -> null
    }

  # forgejo_team_member.occ_acchs will be destroyed
  # (because forgejo_team_member.occ_acchs is not in configuration)
  - resource "forgejo_team_member" "occ_acchs" {
      - team_id = 33 -> null
      - user    = "acchs" -> null
    }

  # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place
  ~ resource "github_repository_ruleset" "cmu-eats_eats_ruleset" {
        id          = "21330218"
        name        = "Default"
        # (6 unchanged attributes hidden)

      ~ bypass_actors {
          ~ actor_id    = 0 -> 161775995
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_team_membership.cmu-maps_krishsax will be destroyed
  # (because github_team_membership.cmu-maps_krishsax is not in configuration)
  - resource "github_team_membership" "cmu-maps_krishsax" {
      - etag     = "W/\"0ca81c85b299f01891dc704fc1e1f11d9b5e73a022c43d28f0593b116aa3098f\"" -> null
      - id       = "10872360:krishsaxena" -> null
      - role     = "maintainer" -> null
      - team_id  = "10872360" -> null
      - username = "krishsaxena" -> null
    }

  # github_team_membership.cmu-maps_siddharthr will be created
  + resource "github_team_membership" "cmu-maps_siddharthr" {
      + etag     = (known after apply)
      + id       = (known after apply)
      + role     = "maintainer"
      + team_id  = "10872360"
      + username = "overthrowing"
    }

  # github_team_membership.cmu-maps_xboxbedrock will be destroyed
  # (because github_team_membership.cmu-maps_xboxbedrock is not in configuration)
  - resource "github_team_membership" "cmu-maps_xboxbedrock" {
      - etag     = "W/\"695105377afc546ab133d841be4a870c042dca48a6172b3388875744e9e1aef9\"" -> null
      - id       = "10872360:xboxbedrock" -> null
      - role     = "maintainer" -> null
      - team_id  = "10872360" -> null
      - username = "xboxbedrock" -> null
    }

  # github_team_membership.occ_acchs will be destroyed
  # (because github_team_membership.occ_acchs is not in configuration)
  - resource "github_team_membership" "occ_acchs" {
      - etag     = "W/\"886b153b13edb3f78b81ca6ef27b40dc0f1fc0536eeef77c854da6ebd9d7342e\"" -> null
      - id       = "19240310:acchs-at-cmu" -> null
      - role     = "member" -> null
      - team_id  = "19240310" -> null
      - username = "acchs-at-cmu" -> null
    }

Plan: 3 to add, 1 to change, 7 to destroy.
╷
│ Error: Unable to read team member
│ 
│ User 'krishsax' in team with ID 13 not found: The target couldn't be found.
╵



Plan Summary

6 projects, 5 with changes, 0 with no changes, 1 failed

  • To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # null_resource.slack_cmu-maps_ch0_krishsax will be destroyed # (because null_resource.slack_cmu-maps_ch0_krishsax is not in configuration) - resource "null_resource" "slack_cmu-maps_ch0_krishsax" { - id = "8867766601843827684" -> null - triggers = { - "channel" = "C05T7QRN4LF" - "user" = "U09DL6NH8FL" } -> null } # null_resource.slack_cmu-maps_ch0_siddharthr will be created + resource "null_resource" "slack_cmu-maps_ch0_siddharthr" { + id = (known after apply) + triggers = { + "channel" = "C05T7QRN4LF" + "user" = "U09DC6MFB5E" } } # null_resource.slack_cmu-maps_ch0_xboxbedrock will be destroyed # (because null_resource.slack_cmu-maps_ch0_xboxbedrock is not in configuration) - resource "null_resource" "slack_cmu-maps_ch0_xboxbedrock" { - id = "2901886759953880290" -> null - triggers = { - "channel" = "C05T7QRN4LF" - "user" = "U0AQRKYLGHK" } -> null } # null_resource.slack_hub_acchs will be destroyed # (because null_resource.slack_hub_acchs is not in configuration) - resource "null_resource" "slack_hub_acchs" { - id = "5543954889836015881" -> null - triggers = { - "channel" = "C096TM8EMS8" - "user" = "U0BVBNX8YGZ" } -> null } # null_resource.slack_hub_siddharthr will be created + resource "null_resource" "slack_hub_siddharthr" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U09DC6MFB5E" } } # null_resource.slack_leads_siddharthr will be created + resource "null_resource" "slack_leads_siddharthr" { + id = (known after apply) + triggers = { + "channel" = "C0AN1M0HDV0" + "user" = "U09DC6MFB5E" } } Plan: 3 to add, 0 to change, 3 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 3 to add, 0 to change, 3 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.27s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # googleplay_app_iam.maps_krishsax will be destroyed # (because googleplay_app_iam.maps_krishsax is not in configuration) - resource "googleplay_app_iam" "maps_krishsax" { - app_id = "com.cmumaps.app" -> null - expanded_permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - user_id = "krishsax@andrew.cmu.edu" -> null } # googleplay_app_iam.maps_siddharthr will be created + resource "googleplay_app_iam" "maps_siddharthr" { + app_id = "com.cmumaps.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_APP_CONTENT", + "CAN_MANAGE_PUBLIC_APKS", + "CAN_MANAGE_PUBLIC_LISTING", + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_REPLY_TO_REVIEWS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "sradhak2@andrew.cmu.edu" } # googleplay_app_iam.maps_xboxbedrock will be destroyed # (because googleplay_app_iam.maps_xboxbedrock is not in configuration) - resource "googleplay_app_iam" "maps_xboxbedrock" { - app_id = "com.cmumaps.app" -> null - expanded_permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - user_id = "spola2@andrew.cmu.edu" -> null } # googleplay_user.krishsax will be destroyed # (because googleplay_user.krishsax is not in configuration) - resource "googleplay_user" "krishsax" { - email = "krishsax@andrew.cmu.edu" -> null - expanded_permissions = [ - "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] -> null - global_permissions = [ - "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] -> null - id = "krishsax@andrew.cmu.edu" -> null - name = "developers/7984360456919420393/users/krishsax@andrew.cmu.edu" -> null } # googleplay_user.siddharthr will be created + resource "googleplay_user" "siddharthr" { + email = "sradhak2@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } Plan: 2 to add, 0 to change, 3 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 2 to add, 0 to change, 3 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.30s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # google_cloud_identity_group_membership.ops_siddharthr will be created + resource "google_cloud_identity_group_membership" "ops_siddharthr" { + create_time = (known after apply) + group = "groups/03fwokq036zsp1f" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "sradhak2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } # google_cloud_identity_group_membership.tech_acchs will be destroyed # (because google_cloud_identity_group_membership.tech_acchs is not in configuration) - resource "google_cloud_identity_group_membership" "tech_acchs" { - create_time = "2026-09-06T04:03:48.051444Z" -> null - group = "groups/04anzqyu1bqbv79" -> null - id = "groups/04anzqyu1bqbv79/memberships/102039638856706939215" -> null - name = "groups/04anzqyu1bqbv79/memberships/102039638856706939215" -> null - type = "USER" -> null - update_time = "2026-09-06T04:03:48.051444Z" -> null - preferred_member_key { - id = "azzhou@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.tech_siddharthr will be created + resource "google_cloud_identity_group_membership" "tech_siddharthr" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "sradhak2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 2 to add, 0 to change, 1 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 2 to add, 0 to change, 1 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.26s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-maps_maps will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-maps_maps" { id = "scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64" ! members = [ - "krishsax", - "spola2", + "sradhak2", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-maps_maps_admins will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-maps_maps_admins" { id = "scottylabs/group-memberships/75db0c7f-7cfb-4d89-b84f-0f50523ed55b" ! members = [ - "krishsax", - "spola2", + "sradhak2", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_occ_occ-main will be updated in-place ! resource "keycloak_group_memberships" "project_occ_occ-main" { id = "scottylabs/group-memberships/a4e98eec-77cf-49f8-8ea6-32cf6332cf36" ! members = [ - "azzhou", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ - "azzhou", + "sradhak2", # (28 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 4 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 4 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.37s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) - destroy OpenTofu will perform the following actions: # discord_member_roles.acchs will be destroyed # (because discord_member_roles.acchs is not in configuration) - resource "discord_member_roles" "acchs" { - id = "1353118209024331948:781930226778767380" -> null - server_id = "1353118209024331948" -> null - user_id = "781930226778767380" -> null - role { - has_role = true -> null - role_id = "1488787088118644848" -> null } - role { - has_role = true -> null - role_id = "1543486517908082728" -> null } } # discord_member_roles.krishsax will be updated in-place ! resource "discord_member_roles" "krishsax" { id = "1353118209024331948:782966391321395240" # (2 unchanged attributes hidden) - role { - has_role = true -> null - role_id = "1526003694708195448" -> null } # (3 unchanged blocks hidden) } # discord_member_roles.siddharthr will be created + resource "discord_member_roles" "siddharthr" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "584841698186756167" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1519792320755662928" } + role { + has_role = true + role_id = "1526003694708195448" } } # discord_member_roles.xboxbedrock will be updated in-place ! resource "discord_member_roles" "xboxbedrock" { id = "1353118209024331948:712398210345205871" # (2 unchanged attributes hidden) - role { - has_role = true -> null - role_id = "1526003694708195448" -> null } # (5 unchanged blocks hidden) } Plan: 1 to add, 2 to change, 1 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 2 to change, 1 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` **Plan Error** <details><summary>Show Output</summary> ``` running '/nix/store/v8n4x020rdfr57a0s3xdq46a3mdd5bk7-opentofu-1.12.5/bin/tofu plan -input=false -refresh -out /var/lib/atlantis/repos/ScottyLabs/governance/188/default/tofu/git/git-default.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/188/default/tofu/git': exit status 1 Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.31s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ data.external.identity_jessica_lr_song: Reading... data.external.identity_rzhu3: Reading... data.external.identity_siddharthr: Reading... data.external.identity_gostmeaper: Reading... data.external.identity_liamwalk: Reading... data.external.identity_jc_965: Reading... data.external.identity_anish: Reading... data.external.identity_siddharthr: Read complete after 1s [id=-] data.external.identity_brightz: Reading... data.external.identity_jc_965: Read complete after 1s [id=-] data.external.identity_nikhilr: Reading... data.external.identity_rzhu3: Read complete after 1s [id=-] data.external.identity_liamwalk: Read complete after 1s [id=-] data.external.identity_gostmeaper: Read complete after 1s [id=-] data.external.identity_kothurvidur: Reading... data.external.identity_jessica_lr_song: Read complete after 1s [id=-] data.external.identity_jw8: Reading... data.external.identity_amzoeee: Reading... data.external.identity_dcrodrig: Reading... data.external.identity_anish: Read complete after 1s [id=-] data.external.identity_kechezon: Reading... data.external.identity_maybe_yiyi: Reading... data.external.identity_brightz: Read complete after 0s [id=-] data.external.identity_dkou: Reading... data.external.identity_zoexue: Reading... data.external.identity_jw8: Read complete after 0s [id=-] data.external.identity_kritdass: Reading... data.external.identity_dcrodrig: Read complete after 0s [id=-] data.external.identity_xboxbedrock: Reading... data.external.identity_nikhilr: Read complete after 0s [id=-] data.external.identity_jiyas: Reading... data.external.identity_amzoeee: Read complete after 0s [id=-] data.external.identity_manrajs: Reading... data.external.identity_jalenluorion: Reading... data.external.identity_kechezon: Read complete after 0s [id=-] data.external.identity_tentype: Reading... data.external.identity_maybe_yiyi: Read complete after 0s [id=-] data.external.identity_thesuperrl: Reading... data.external.identity_kothurvidur: Read complete after 0s [id=-] data.external.identity_emmawong: Reading... data.external.identity_dkou: Read complete after 0s [id=-] data.external.identity_rhsen: Reading... data.external.identity_zoexue: Read complete after 0s [id=-] data.external.identity_etashjha: Reading... data.external.identity_kritdass: Read complete after 0s [id=-] data.external.identity_krishsax: Reading... data.external.identity_jiyas: Read complete after 0s [id=-] data.external.identity_riztofu: Reading... data.external.identity_manrajs: Read complete after 0s [id=-] random_password.kennel_webhook_secret: Refreshing state... [id=none] github_repository.terrier_terrier: Refreshing state... [id=terrier] data.external.identity_jalenluorion: Read complete after 0s [id=-] github_team_membership.cmu-maps_xboxbedrock: Refreshing state... [id=10872360:xboxbedrock] data.external.identity_xboxbedrock: Read complete after 0s [id=-] github_team_membership.occ_acchs: Refreshing state... [id=19240310:acchs-at-cmu] data.external.identity_tentype: Read complete after 0s [id=-] github_team.cmu-courses: Refreshing state... [id=3643907] data.external.identity_thesuperrl: Read complete after 0s [id=-] github_repository.slai_cmugpt_agent: Refreshing state... [id=cmugpt-agent] data.external.identity_emmawong: Read complete after 0s [id=-] data.external.identity_rhsen: Read complete after 1s [id=-] github_team.cmu-eats: Refreshing state... [id=16807600] github_team.cbp: Refreshing state... [id=17170099] github_repository.cbp_groupme_mirror: Refreshing state... [id=groupme-mirror] data.external.identity_etashjha: Read complete after 1s [id=-] github_repository.devops_collie: Refreshing state... [id=collie] data.external.identity_krishsax: Read complete after 1s [id=-] data.external.identity_riztofu: Read complete after 1s [id=-] github_repository.cmu-housing_housing: Refreshing state... [id=housing] github_team.uia: Refreshing state... [id=17529763] github_team.cmu-housing: Refreshing state... [id=17937919] github_team.occ: Refreshing state... [id=19240310] github_team.cmu-maps: Refreshing state... [id=10872360] github_repository.devops_infrastructure: Refreshing state... [id=infrastructure] github_team.cmu-cal: Refreshing state... [id=15669112] github_team.tartan-vote: Refreshing state... [id=17701833] github_team.slai: Refreshing state... [id=17571328] github_repository.cmu-insta_insta: Refreshing state... [id=insta] github_repository.cmu-courses_internet_archive: Refreshing state... [id=internet-archive] github_repository.tartan-vote_tartan_vote: Refreshing state... [id=tartan-vote] github_team_membership.cmu-maps_krishsax: Refreshing state... [id=10872360:krishsaxena] github_team.cmu-insta: Refreshing state... [id=19240311] github_repository.devops_governance: Refreshing state... [id=governance] github_repository.slai_cmugpt_surface: Refreshing state... [id=cmugpt-surface] github_repository.devops_link_shortener: Refreshing state... [id=link-shortener] github_repository.cmu-eats_eats: Refreshing state... [id=eats] github_team.lost-and-found: Refreshing state... [id=4088608] github_repository.slai_cmugpt_sms_surface: Refreshing state... [id=cmugpt-sms-surface] github_repository.slai_mcp_server: Refreshing state... [id=mcp-server] github_repository.uia_scottylabs_org: Refreshing state... [id=scottylabs-org] github_repository.cmu-eats_dining_api: Refreshing state... [id=dining-api] github_team.terrier: Refreshing state... [id=12028577] github_repository.cmu-courses_courses: Refreshing state... [id=courses] github_repository.devops_kennel: Refreshing state... [id=kennel] github_repository.uia_components: Refreshing state... [id=components] github_repository.cmu-cal_cal: Refreshing state... [id=cal] github_repository.cbp_bus_sign: Refreshing state... [id=bus-sign] github_repository.devops_documentation: Refreshing state... [id=documentation] github_repository.quest_quest: Refreshing state... [id=quest] github_repository.lost-and-found_lost_and_found_v2: Refreshing state... [id=lost-and-found-v2] github_repository.occ_occ: Refreshing state... [id=occ] github_repository.cbp_discord_verify: Refreshing state... [id=discord-verify] github_team.quest: Refreshing state... [id=17330197] github_team.devops: Refreshing state... [id=14706218] github_repository.cbp_dalmatian: Refreshing state... [id=dalmatian] github_repository.cmu-maps_maps: Refreshing state... [id=maps] forgejo_repository.cmu-housing_housing: Refreshing state... [name=housing] forgejo_team_member.occ_acchs: Refreshing state... forgejo_repository.devops_infrastructure: Refreshing state... [name=infrastructure] forgejo_team_member.cmu-maps_krishsax: Refreshing state... forgejo_team_member.cmu-maps_xboxbedrock: Refreshing state... forgejo_repository.cmu-maps_maps: Refreshing state... [name=maps] forgejo_repository.cmu-eats_eats: Refreshing state... [name=eats] forgejo_repository.occ_occ: Refreshing state... [name=occ] forgejo_repository.cmu-insta_insta: Refreshing state... [name=insta] forgejo_repository.devops_kennel: Refreshing state... [name=kennel] forgejo_repository.cbp_groupme_mirror: Refreshing state... [name=groupme-mirror] forgejo_repository.lost-and-found_lost_and_found_v2: Refreshing state... [name=lost-and-found-v2] forgejo_repository.slai_mcp_server: Refreshing state... [name=mcp-server] forgejo_team_member.maps_leads_krishsax: Refreshing state... forgejo_repository.slai_cmugpt_sms_surface: Refreshing state... [name=cmugpt-sms-surface] forgejo_repository.tartan-vote_tartan_vote: Refreshing state... [name=tartan-vote] forgejo_repository.devops_documentation: Refreshing state... [name=documentation] forgejo_team_member.maps_leads_xboxbedrock: Refreshing state... forgejo_repository.uia_components: Refreshing state... [name=components] forgejo_repository.cmu-cal_cal: Refreshing state... [name=cal] forgejo_repository.cbp_dalmatian: Refreshing state... [name=dalmatian] forgejo_repository.quest_quest: Refreshing state... [name=quest] forgejo_repository.cbp_discord_verify: Refreshing state... [name=discord-verify] forgejo_repository.cmu-courses_internet_archive: Refreshing state... [name=internet-archive] data.forgejo_organization.this: Reading... forgejo_repository.devops_link_shortener: Refreshing state... [name=link-shortener] data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs] forgejo_repository.devops_collie: Refreshing state... [name=collie] forgejo_repository.devops_governance: Refreshing state... [name=governance] forgejo_repository.cmu-courses_courses: Refreshing state... [name=courses] forgejo_repository.cbp_bus_sign: Refreshing state... [name=bus-sign] forgejo_repository.terrier_terrier: Refreshing state... [name=terrier] forgejo_repository.slai_cmugpt_surface: Refreshing state... [name=cmugpt-surface] forgejo_repository.uia_scottylabs_org: Refreshing state... [name=scottylabs-org] forgejo_repository.slai_cmugpt_agent: Refreshing state... [name=cmugpt-agent] forgejo_repository.cmu-eats_dining_api: Refreshing state... [name=dining-api] github_actions_repository_permissions.terrier_terrier_actions: Refreshing state... [id=terrier] github_team_membership.cmu-courses_anish: Refreshing state... [id=3643907:ap-1] github_team_membership.cmu-courses_etashjha: Refreshing state... [id=3643907:etashj] github_actions_repository_permissions.slai_cmugpt_agent_actions: Refreshing state... [id=cmugpt-agent] github_team_membership.cmu-eats_anish: Refreshing state... [id=16807600:ap-1] github_team_membership.cbp_liamwalk: Refreshing state... [id=17170099:pandaaured] github_team_membership.cbp_maybe_yiyi: Refreshing state... [id=17170099:maybe-yiyi] github_team_membership.cbp_tentype: Refreshing state... [id=17170099:tentype] github_actions_repository_permissions.cbp_groupme_mirror_actions: Refreshing state... [id=groupme-mirror] github_actions_repository_permissions.devops_collie_actions: Refreshing state... [id=collie] github_team_membership.cmu-housing_emmawong: Refreshing state... [id=17937919:emw868] github_team_membership.cmu-housing_rhsen: Refreshing state... [id=17937919:rohansen08] github_team_membership.cmu-housing_gostmeaper: Refreshing state... [id=17937919:gostmeaper] github_team_membership.cmu-housing_manrajs: Refreshing state... [id=17937919:manraj10] github_team_membership.cmu-housing_nikhilr: Refreshing state... [id=17937919:ecstaticpilot] github_team_membership.cmu-housing_tentype: Refreshing state... [id=17937919:tentype] github_team_membership.cmu-housing_zoexue: Refreshing state... [id=17937919:o-zox-o] github_team_membership.cmu-housing_xboxbedrock: Refreshing state... [id=17937919:xboxbedrock] github_team_membership.cmu-housing_amzoeee: Refreshing state... [id=17937919:amzoeee] github_team_membership.cmu-housing_jessica_lr_song: Refreshing state... [id=17937919:jessica-lr-song] github_team_membership.occ_anish: Refreshing state... [id=19240310:ap-1] github_team_membership.occ_dcrodrig: Refreshing state... [id=19240310:newhacker1746] github_team_membership.cmu-maps_anish: Refreshing state... [id=10872360:ap-1] github_team_membership.cmu-cal_jalenluorion: Refreshing state... [id=15669112:jalenluorion] github_team_membership.tartan-vote_maybe_yiyi: Refreshing state... [id=17701833:maybe-yiyi] github_team_membership.tartan-vote_riztofu: Refreshing state... [id=17701833:riztofu] github_team_membership.slai_rzhu3: Refreshing state... [id=17571328:richard-31415] github_team_membership.slai_krishsax: Refreshing state... [id=17571328:krishsaxena] github_team_membership.slai_jc_965: Refreshing state... [id=17571328:jc-965] github_team_membership.slai_jiyas: Refreshing state... [id=17571328:jiya2301] github_actions_repository_permissions.cmu-courses_internet_archive_actions: Refreshing state... [id=internet-archive] github_actions_repository_permissions.cmu-housing_housing_actions: Refreshing state... [id=housing] github_team_membership.cmu-insta_etashjha: Refreshing state... [id=19240311:etashj] github_actions_repository_permissions.devops_infrastructure_actions: Refreshing state... [id=infrastructure] github_actions_repository_permissions.slai_cmugpt_surface_actions: Refreshing state... [id=cmugpt-surface] github_actions_repository_permissions.cmu-insta_insta_actions: Refreshing state... [id=insta] github_actions_repository_permissions.tartan-vote_tartan_vote_actions: Refreshing state... [id=tartan-vote] github_actions_repository_permissions.slai_cmugpt_sms_surface_actions: Refreshing state... [id=cmugpt-sms-surface] github_actions_repository_permissions.devops_governance_actions: Refreshing state... [id=governance] github_actions_repository_permissions.devops_link_shortener_actions: Refreshing state... [id=link-shortener] github_team_membership.terrier_dkou: Refreshing state... [id=12028577:sharkree] github_team_membership.terrier_kritdass: Refreshing state... [id=12028577:kritdass] github_team_membership.terrier_etashjha: Refreshing state... [id=12028577:etashj] github_team_membership.terrier_kothurvidur: Refreshing state... [id=12028577:kothurvidur] github_team_membership.terrier_anish: Refreshing state... [id=12028577:ap-1] github_actions_repository_permissions.cmu-eats_eats_actions: Refreshing state... [id=eats] github_actions_repository_permissions.cmu-courses_courses_actions: Refreshing state... [id=courses] github_actions_repository_permissions.uia_components_actions: Refreshing state... [id=components] github_actions_repository_permissions.slai_mcp_server_actions: Refreshing state... [id=mcp-server] github_actions_repository_permissions.uia_scottylabs_org_actions: Refreshing state... [id=scottylabs-org] github_actions_repository_permissions.cmu-eats_dining_api_actions: Refreshing state... [id=dining-api] github_actions_repository_permissions.devops_documentation_actions: Refreshing state... [id=documentation] github_actions_repository_permissions.lost-and-found_lost_and_found_v2_actions: Refreshing state... [id=lost-and-found-v2] github_actions_repository_permissions.devops_kennel_actions: Refreshing state... [id=kennel] restapi_object.cmu-housing_housing_push_mirror: Refreshing state... [id=remote_mirror_uoV2HG4aiOT] restapi_object.cmu-housing_housing_docs_webhook: Refreshing state... [id=15] restapi_object.devops_infrastructure_push_mirror: Refreshing state... [id=remote_mirror_oYtFpKVzTaZ] restapi_object.cmu-maps_maps_push_mirror: Refreshing state... [id=remote_mirror_e4_FUnvoslJ] restapi_object.cmu-eats_eats_push_mirror: Refreshing state... [id=remote_mirror_sjYwMA6NPgh] restapi_object.occ_occ_push_mirror: Refreshing state... [id=remote_mirror_moBLq1YG1qz] restapi_object.cmu-insta_insta_push_mirror: Refreshing state... [id=remote_mirror_891nu6XQ1us] restapi_object.devops_kennel_push_mirror: Refreshing state... [id=remote_mirror_cT9psJkhTQ4] restapi_object.cbp_groupme_mirror_push_mirror: Refreshing state... [id=remote_mirror_U61HJHB9VGE] restapi_object.lost-and-found_lost_and_found_v2_push_mirror: Refreshing state... [id=remote_mirror_n_UZBFU-CY4] restapi_object.slai_mcp_server_push_mirror: Refreshing state... [id=remote_mirror_Tqh2y0R_vDp] restapi_object.slai_cmugpt_sms_surface_push_mirror: Refreshing state... [id=remote_mirror__fkBkASN2jA] restapi_object.tartan-vote_tartan_vote_push_mirror: Refreshing state... [id=remote_mirror_EUFq4sFSnH5] restapi_object.tartan-vote_tartan_vote_docs_webhook: Refreshing state... [id=25] restapi_object.devops_documentation_push_mirror: Refreshing state... [id=remote_mirror_izmsqaNB6Wx] github_actions_repository_permissions.cbp_discord_verify_actions: Refreshing state... [id=discord-verify] restapi_object.uia_components_push_mirror: Refreshing state... [id=remote_mirror_Kwwa8TDnTgc] restapi_object.cmu-cal_cal_push_mirror: Refreshing state... [id=remote_mirror_CO243uIONoq] restapi_object.cbp_dalmatian_push_mirror: Refreshing state... [id=remote_mirror_qzlCpLK153F] restapi_object.cbp_dalmatian_docs_webhook: Refreshing state... [id=7] restapi_object.quest_quest_push_mirror: Refreshing state... [id=remote_mirror_U9d9BPAon29] restapi_object.cmu-courses_internet_archive_push_mirror: Refreshing state... [id=remote_mirror_58X__Yfs0bR] restapi_object.cbp_discord_verify_push_mirror: Refreshing state... [id=remote_mirror_3MV7Up6cY1X] github_team_membership.quest_anish: Refreshing state... [id=17330197:ap-1] github_team_membership.quest_kechezon: Refreshing state... [id=17330197:kechezon] github_team_membership.quest_brightz: Refreshing state... [id=17330197:bright-zheng] github_team_membership.quest_xboxbedrock: Refreshing state... [id=17330197:xboxbedrock] github_team_membership.quest_jw8: Refreshing state... [id=17330197:jeffreyyijiwang] forgejo_team.terrier: Refreshing state... [name=terrier] forgejo_team.lost-and-found: Refreshing state... [name=lost-and-found] forgejo_team.tartan-vote_leads: Refreshing state... [name=tartan-vote-leads] forgejo_team.dalmatian_leads: Refreshing state... [name=dalmatian-leads] forgejo_team.insta_leads: Refreshing state... [name=insta-leads] forgejo_team.cmu-insta: Refreshing state... [name=cmu-insta] forgejo_team.cmu-maps: Refreshing state... [name=cmu-maps] forgejo_team.devops: Refreshing state... [name=devops] forgejo_team.maps_leads: Refreshing state... [name=maps-leads] forgejo_team.eats_leads: Refreshing state... [name=eats-leads] forgejo_team.tartan-vote: Refreshing state... [name=tartan-vote] forgejo_team.groupme-mirror_leads: Refreshing state... [name=groupme-mirror-leads] forgejo_team.cmu-eats: Refreshing state... [name=cmu-eats] forgejo_team.bus-sign_leads: Refreshing state... [name=bus-sign-leads] forgejo_team.cbp: Refreshing state... [name=cbp] forgejo_team.terrier_leads: Refreshing state... [name=terrier-leads] forgejo_team.discord-verify_leads: Refreshing state... [name=discord-verify-leads] forgejo_team.quest: Refreshing state... [name=quest] forgejo_team.occ: Refreshing state... [name=occ] forgejo_team.cmugpt-surface_leads: Refreshing state... [name=cmugpt-surface-leads] forgejo_team.documentation_leads: Refreshing state... [name=documentation-leads] forgejo_team.courses_leads: Refreshing state... [name=courses-leads] forgejo_team.devops_leads: Refreshing state... [name=devops-leads] forgejo_team.cmugpt-mcp-server_leads: Refreshing state... [name=cmugpt-mcp-server-leads] forgejo_team.link-shortener_leads: Refreshing state... [name=link-shortener-leads] forgejo_team.housing_leads: Refreshing state... [name=housing-leads] forgejo_team.occ-main_leads: Refreshing state... [name=occ-main-leads] forgejo_team.slai: Refreshing state... [name=slai] forgejo_team.cmu-courses: Refreshing state... [name=cmu-courses] forgejo_team.cmu-courses_leads: Refreshing state... [name=cmu-courses-leads] forgejo_team.cal_leads: Refreshing state... [name=cal-leads] forgejo_team.cmugpt-sms-surface_leads: Refreshing state... [name=cmugpt-sms-surface-leads] forgejo_team.uia: Refreshing state... [name=uia] forgejo_team.quest_leads: Refreshing state... [name=quest-leads] forgejo_team.cmugpt-agent_leads: Refreshing state... [name=cmugpt-agent-leads] forgejo_team.cmu-housing: Refreshing state... [name=cmu-housing] forgejo_team.cmu-cal: Refreshing state... [name=cmu-cal] restapi_object.devops_link_shortener_push_mirror: Refreshing state... [id=remote_mirror_cp8daMpt82v] restapi_object.devops_collie_push_mirror: Refreshing state... [id=remote_mirror_ljvVejNFXaS] restapi_object.devops_governance_push_mirror: Refreshing state... [id=remote_mirror_lt9JzkX1jMq] restapi_object.cmu-courses_courses_push_mirror: Refreshing state... [id=remote_mirror_WMR9yQwKFDF] restapi_object.cbp_bus_sign_push_mirror: Refreshing state... [id=remote_mirror_MkKxQ2vA9pc] restapi_object.terrier_terrier_push_mirror: Refreshing state... [id=remote_mirror_YopDgZksZv5] restapi_object.slai_cmugpt_surface_push_mirror: Refreshing state... [id=remote_mirror_cPTQN6S5SpB] github_actions_repository_permissions.cmu-cal_cal_actions: Refreshing state... [id=cal] restapi_object.cmu-courses_courses_kennel_webhook: Refreshing state... [id=11] restapi_object.cmu-eats_eats_kennel_webhook: Refreshing state... [id=31] restapi_object.cbp_discord_verify_kennel_webhook: Refreshing state... [id=18] restapi_object.uia_components_kennel_webhook: Refreshing state... [id=27] restapi_object.slai_cmugpt_surface_kennel_webhook: Refreshing state... [id=23] restapi_object.cmu-housing_housing_kennel_webhook: Refreshing state... [id=16] restapi_object.devops_kennel_kennel_webhook: Refreshing state... [id=14] restapi_object.terrier_terrier_kennel_webhook: Refreshing state... [id=12] restapi_object.cmu-maps_maps_kennel_webhook: Refreshing state... [id=9] restapi_object.cmu-cal_cal_kennel_webhook: Refreshing state... [id=20] restapi_object.cbp_bus_sign_kennel_webhook: Refreshing state... [id=21] restapi_object.tartan-vote_tartan_vote_kennel_webhook: Refreshing state... [id=24] restapi_object.devops_link_shortener_kennel_webhook: Refreshing state... [id=8] restapi_object.devops_infrastructure_kennel_webhook: Refreshing state... [id=26] restapi_object.devops_governance_kennel_webhook: Refreshing state... [id=22] restapi_object.slai_cmugpt_sms_surface_kennel_webhook: Refreshing state... [id=19] restapi_object.quest_quest_kennel_webhook: Refreshing state... [id=10] restapi_object.slai_cmugpt_agent_push_mirror: Refreshing state... [id=remote_mirror_kNY6rVcxdyr] restapi_object.slai_cmugpt_agent_kennel_webhook: Refreshing state... [id=28] restapi_object.slai_mcp_server_kennel_webhook: Refreshing state... [id=13] restapi_object.devops_collie_kennel_webhook: Refreshing state... [id=17] restapi_object.cbp_dalmatian_kennel_webhook: Refreshing state... [id=6] restapi_object.cmu-eats_dining_api_push_mirror: Refreshing state... [id=remote_mirror_a4MsQ5rO1_Z] restapi_object.uia_scottylabs_org_push_mirror: Refreshing state... [id=remote_mirror_h6yN6QdS1lD] restapi_object.uia_scottylabs_org_kennel_webhook: Refreshing state... [id=29] github_team_membership.devops_anish: Refreshing state... [id=14706218:ap-1] github_team_membership.devops_xboxbedrock: Refreshing state... [id=14706218:xboxbedrock] github_team_membership.devops_thesuperrl: Refreshing state... [id=14706218:thesuperrl] github_actions_repository_permissions.cbp_bus_sign_actions: Refreshing state... [id=bus-sign] github_actions_repository_permissions.quest_quest_actions: Refreshing state... [id=quest] github_actions_repository_permissions.occ_occ_actions: Refreshing state... [id=occ] github_actions_repository_permissions.cbp_dalmatian_actions: Refreshing state... [id=dalmatian] github_actions_repository_permissions.cmu-maps_maps_actions: Refreshing state... [id=maps] github_repository_deploy_key.cmu-housing_housing_mirror_deploy_key: Refreshing state... [id=housing:159665369] github_repository_deploy_key.devops_infrastructure_mirror_deploy_key: Refreshing state... [id=infrastructure:159665343] github_repository_deploy_key.cmu-maps_maps_mirror_deploy_key: Refreshing state... [id=maps:159665325] github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Refreshing state... [id=eats:161775995] github_repository_deploy_key.occ_occ_mirror_deploy_key: Refreshing state... [id=occ:161720242] github_repository_deploy_key.cmu-insta_insta_mirror_deploy_key: Refreshing state... [id=insta:161720245] github_repository_deploy_key.devops_kennel_mirror_deploy_key: Refreshing state... [id=kennel:159665328] github_repository_deploy_key.cbp_groupme_mirror_mirror_deploy_key: Refreshing state... [id=groupme-mirror:159665366] github_repository_deploy_key.lost-and-found_lost_and_found_v2_mirror_deploy_key: Refreshing state... [id=lost-and-found-v2:159665337] github_repository_deploy_key.slai_mcp_server_mirror_deploy_key: Refreshing state... [id=mcp-server:159665358] github_repository_deploy_key.slai_cmugpt_sms_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-sms-surface:159665365] github_repository_deploy_key.tartan-vote_tartan_vote_mirror_deploy_key: Refreshing state... [id=tartan-vote:159665350] github_repository_deploy_key.devops_documentation_mirror_deploy_key: Refreshing state... [id=documentation:159665362] github_repository_deploy_key.uia_components_mirror_deploy_key: Refreshing state... [id=components:159665353] github_repository_deploy_key.cbp_dalmatian_mirror_deploy_key: Refreshing state... [id=dalmatian:159665367] github_repository_deploy_key.cmu-cal_cal_mirror_deploy_key: Refreshing state... [id=cal:159665344] github_repository_deploy_key.quest_quest_mirror_deploy_key: Refreshing state... [id=quest:159665323] github_repository_deploy_key.cmu-courses_internet_archive_mirror_deploy_key: Refreshing state... [id=internet-archive:159665356] github_repository_deploy_key.cbp_discord_verify_mirror_deploy_key: Refreshing state... [id=discord-verify:159665340] forgejo_team_member.terrier_kothurvidur: Refreshing state... forgejo_team_member.terrier_anish: Refreshing state... forgejo_team_member.terrier_kritdass: Refreshing state... forgejo_team_repository.terrier_terrier: Refreshing state... forgejo_team_member.terrier_etashjha: Refreshing state... forgejo_team_member.terrier_dkou: Refreshing state... forgejo_team_repository.lost-and-found_lost_and_found_v2: Refreshing state... forgejo_team_member.tartan-vote_leads_maybe_yiyi: Refreshing state... forgejo_team_member.dalmatian_leads_maybe_yiyi: Refreshing state... forgejo_team_member.dalmatian_leads_tentype: Refreshing state... forgejo_team_member.insta_leads_etashjha: Refreshing state... forgejo_team_member.cmu-insta_etashjha: Refreshing state... forgejo_team_repository.cmu-insta_insta: Refreshing state... forgejo_team_repository.cmu-maps_maps: Refreshing state... forgejo_team_member.cmu-maps_anish: Refreshing state... forgejo_team_repository.devops_governance: Refreshing state... forgejo_team_repository.devops_collie: Refreshing state... forgejo_team_repository.devops_link_shortener: Refreshing state... forgejo_team_member.devops_xboxbedrock: Refreshing state... forgejo_team_repository.devops_documentation: Refreshing state... forgejo_team_member.devops_thesuperrl: Refreshing state... forgejo_team_repository.devops_infrastructure: Refreshing state... forgejo_team_member.devops_anish: Refreshing state... forgejo_team_repository.devops_kennel: Refreshing state... forgejo_team_member.maps_leads_anish: Refreshing state... forgejo_team_member.tartan-vote_riztofu: Refreshing state... forgejo_team_member.eats_leads_anish: Refreshing state... forgejo_team_member.tartan-vote_maybe_yiyi: Refreshing state... forgejo_team_repository.tartan-vote_tartan_vote: Refreshing state... forgejo_team_member.groupme-mirror_leads_maybe_yiyi: Refreshing state... forgejo_team_member.groupme-mirror_leads_tentype: Refreshing state... forgejo_team_repository.cmu-eats_eats: Refreshing state... forgejo_team_member.cmu-eats_anish: Refreshing state... forgejo_team_repository.cmu-eats_dining_api: Refreshing state... forgejo_team_member.bus-sign_leads_maybe_yiyi: Refreshing state... forgejo_team_member.bus-sign_leads_tentype: Refreshing state... forgejo_team_member.cbp_tentype: Refreshing state... forgejo_team_repository.cbp_discord_verify: Refreshing state... forgejo_team_repository.cbp_dalmatian: Refreshing state... forgejo_team_repository.cbp_groupme_mirror: Refreshing state... forgejo_team_member.cbp_maybe_yiyi: Refreshing state... forgejo_team_repository.cbp_bus_sign: Refreshing state... forgejo_team_member.cbp_liamwalk: Refreshing state... forgejo_team_member.terrier_leads_kritdass: Refreshing state... forgejo_team_member.terrier_leads_anish: Refreshing state... forgejo_team_member.discord-verify_leads_tentype: Refreshing state... forgejo_team_member.discord-verify_leads_maybe_yiyi: Refreshing state... forgejo_team_repository.quest_quest: Refreshing state... forgejo_team_member.quest_brightz: Refreshing state... forgejo_team_member.quest_jw8: Refreshing state... forgejo_team_member.quest_kechezon: Refreshing state... forgejo_team_member.quest_anish: Refreshing state... forgejo_team_member.quest_xboxbedrock: Refreshing state... forgejo_team_member.occ_anish: Refreshing state... forgejo_team_repository.occ_occ: Refreshing state... forgejo_team_member.occ_dcrodrig: Refreshing state... forgejo_team_member.cmugpt-surface_leads_krishsax: Refreshing state... forgejo_team_member.documentation_leads_thesuperrl: Refreshing state... forgejo_team_member.documentation_leads_anish: Refreshing state... forgejo_team_member.devops_leads_anish: Refreshing state... forgejo_team_member.devops_leads_thesuperrl: Refreshing state... forgejo_team_member.courses_leads_anish: Refreshing state... forgejo_team_member.courses_leads_etashjha: Refreshing state... forgejo_team_member.cmugpt-mcp-server_leads_krishsax: Refreshing state... forgejo_team_member.link-shortener_leads_anish: Refreshing state... forgejo_team_member.link-shortener_leads_thesuperrl: Refreshing state... forgejo_team_member.housing_leads_tentype: Refreshing state... forgejo_team_member.housing_leads_emmawong: Refreshing state... forgejo_team_member.housing_leads_nikhilr: Refreshing state... forgejo_team_member.housing_leads_gostmeaper: Refreshing state... forgejo_team_member.occ-main_leads_dcrodrig: Refreshing state... forgejo_team_member.occ-main_leads_anish: Refreshing state... forgejo_team_member.slai_jiyas: Refreshing state... forgejo_team_repository.slai_mcp_server: Refreshing state... forgejo_team_repository.slai_cmugpt_agent: Refreshing state... forgejo_team_repository.slai_cmugpt_surface: Refreshing state... forgejo_team_member.slai_krishsax: Refreshing state... forgejo_team_member.slai_jc_965: Refreshing state... forgejo_team_repository.slai_cmugpt_sms_surface: Refreshing state... forgejo_team_member.slai_rzhu3: Refreshing state... forgejo_team_member.cmu-courses_leads_etashjha: Refreshing state... forgejo_team_member.cmu-courses_leads_anish: Refreshing state... forgejo_team_member.cmu-courses_anish: Refreshing state... forgejo_team_repository.cmu-courses_internet_archive: Refreshing state... forgejo_team_repository.cmu-courses_courses: Refreshing state... forgejo_team_member.cmu-courses_etashjha: Refreshing state... forgejo_team_member.cal_leads_jalenluorion: Refreshing state... forgejo_team_member.cmugpt-sms-surface_leads_krishsax: Refreshing state... forgejo_team_member.quest_leads_anish: Refreshing state... forgejo_team_member.quest_leads_xboxbedrock: Refreshing state... forgejo_team_repository.uia_scottylabs_org: Refreshing state... forgejo_team_repository.cmu-housing_housing: Refreshing state... forgejo_team_repository.uia_components: Refreshing state... forgejo_team_member.cmu-housing_tentype: Refreshing state... forgejo_team_member.cmu-housing_emmawong: Refreshing state... forgejo_team_member.cmu-housing_rhsen: Refreshing state... forgejo_team_member.cmu-housing_xboxbedrock: Refreshing state... forgejo_team_member.cmu-housing_manrajs: Refreshing state... forgejo_team_member.cmu-housing_zoexue: Refreshing state... forgejo_team_member.cmu-housing_nikhilr: Refreshing state... forgejo_team_member.cmu-housing_jessica_lr_song: Refreshing state... forgejo_team_member.cmu-housing_gostmeaper: Refreshing state... forgejo_team_member.cmu-housing_amzoeee: Refreshing state... forgejo_team_repository.cmu-cal_cal: Refreshing state... forgejo_team_member.cmu-cal_jalenluorion: Refreshing state... forgejo_team_member.cmugpt-agent_leads_krishsax: Refreshing state... github_repository_deploy_key.devops_link_shortener_mirror_deploy_key: Refreshing state... [id=link-shortener:159665345] github_repository_deploy_key.devops_collie_mirror_deploy_key: Refreshing state... [id=collie:159665332] github_repository_deploy_key.devops_governance_mirror_deploy_key: Refreshing state... [id=governance:159665346] github_repository_deploy_key.cmu-courses_courses_mirror_deploy_key: Refreshing state... [id=courses:159665327] github_repository_deploy_key.terrier_terrier_mirror_deploy_key: Refreshing state... [id=terrier:159665339] github_repository_deploy_key.cbp_bus_sign_mirror_deploy_key: Refreshing state... [id=bus-sign:159665342] github_repository_deploy_key.slai_cmugpt_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-surface:159665348] github_repository_deploy_key.slai_cmugpt_agent_mirror_deploy_key: Refreshing state... [id=cmugpt-agent:159665360] github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Refreshing state... [id=dining-api:161772815] github_repository_deploy_key.uia_scottylabs_org_mirror_deploy_key: Refreshing state... [id=scottylabs-org:159665361] github_repository_ruleset.cmu-housing_housing_ruleset: Refreshing state... [id=17421604] github_repository_ruleset.devops_infrastructure_ruleset: Refreshing state... [id=11275118] github_repository_ruleset.cmu-maps_maps_ruleset: Refreshing state... [id=18864961] github_repository_ruleset.cmu-eats_eats_ruleset: Refreshing state... [id=21330218] github_repository_ruleset.occ_occ_ruleset: Refreshing state... [id=21841634] github_repository_ruleset.cmu-insta_insta_ruleset: Refreshing state... [id=21841636] github_repository_ruleset.devops_kennel_ruleset: Refreshing state... [id=13193453] github_repository_ruleset.cbp_groupme_mirror_ruleset: Refreshing state... [id=12503296] github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Refreshing state... [id=18920618] github_repository_ruleset.slai_mcp_server_ruleset: Refreshing state... [id=17082154] forgejo_team_repository.terrier_leads_terrier: Refreshing state... github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Refreshing state... [id=17653622] forgejo_team_repository.insta_leads_insta: Refreshing state... forgejo_team_repository.maps_leads_maps: Refreshing state... forgejo_team_repository.devops_leads_governance: Refreshing state... forgejo_team_repository.devops_leads_collie: Refreshing state... forgejo_team_repository.link-shortener_leads_link_shortener: Refreshing state... forgejo_team_repository.documentation_leads_documentation: Refreshing state... forgejo_team_repository.devops_leads_infrastructure: Refreshing state... github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Refreshing state... [id=16848966] forgejo_team_repository.devops_leads_kennel: Refreshing state... forgejo_team_repository.tartan-vote_leads_tartan_vote: Refreshing state... forgejo_team_repository.eats_leads_eats: Refreshing state... forgejo_team_repository.eats_leads_dining_api: Refreshing state... forgejo_team_repository.discord-verify_leads_discord_verify: Refreshing state... forgejo_team_repository.dalmatian_leads_dalmatian: Refreshing state... forgejo_team_repository.groupme-mirror_leads_groupme_mirror: Refreshing state... forgejo_team_repository.bus-sign_leads_bus_sign: Refreshing state... github_repository_ruleset.devops_documentation_ruleset: Refreshing state... [id=17079368] forgejo_team_repository.quest_leads_quest: Refreshing state... forgejo_team_repository.occ-main_leads_occ: Refreshing state... github_repository_ruleset.uia_components_ruleset: Refreshing state... [id=16366085] forgejo_team_repository.cmugpt-agent_leads_cmugpt_agent: Refreshing state... forgejo_team_repository.cmugpt-surface_leads_cmugpt_surface: Refreshing state... forgejo_team_repository.cmugpt-mcp-server_leads_mcp_server: Refreshing state... forgejo_team_repository.cmugpt-sms-surface_leads_cmugpt_sms_surface: Refreshing state... forgejo_team_repository.cmu-courses_leads_internet_archive: Refreshing state... forgejo_team_repository.courses_leads_courses: Refreshing state... forgejo_team_repository.housing_leads_housing: Refreshing state... forgejo_team_repository.cal_leads_cal: Refreshing state... github_repository_ruleset.cbp_dalmatian_ruleset: Refreshing state... [id=17238915] github_repository_ruleset.cmu-cal_cal_ruleset: Refreshing state... [id=18919281] github_repository_ruleset.quest_quest_ruleset: Refreshing state... [id=15750052] github_repository_ruleset.cmu-courses_internet_archive_ruleset: Refreshing state... [id=11988074] github_repository_ruleset.cbp_discord_verify_ruleset: Refreshing state... [id=11449154] github_repository_ruleset.devops_governance_ruleset: Refreshing state... [id=14488792] github_repository_ruleset.devops_collie_ruleset: Refreshing state... [id=18865204] github_repository_ruleset.devops_link_shortener_ruleset: Refreshing state... [id=17911647] github_repository_ruleset.cmu-courses_courses_ruleset: Refreshing state... [id=15855161] github_repository_ruleset.terrier_terrier_ruleset: Refreshing state... [id=12671003] github_repository_ruleset.cbp_bus_sign_ruleset: Refreshing state... [id=15236550] github_repository_ruleset.slai_cmugpt_surface_ruleset: Refreshing state... [id=16476662] github_repository_ruleset.slai_cmugpt_agent_ruleset: Refreshing state... [id=16476661] github_repository_ruleset.cmu-eats_dining_api_ruleset: Refreshing state... [id=21879626] github_repository_ruleset.uia_scottylabs_org_ruleset: Refreshing state... [id=18112214] OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ~ update in-place (current -> planned) - destroy OpenTofu planned the following actions, but then encountered a problem: # forgejo_team_member.cmu-maps_siddharthr will be created + resource "forgejo_team_member" "cmu-maps_siddharthr" { + team_id = 13 + user = "sradhak2" } # forgejo_team_member.cmu-maps_xboxbedrock will be destroyed # (because forgejo_team_member.cmu-maps_xboxbedrock is not in configuration) - resource "forgejo_team_member" "cmu-maps_xboxbedrock" { - team_id = 13 -> null - user = "xboxbedrock" -> null } # forgejo_team_member.maps_leads_krishsax will be destroyed # (because forgejo_team_member.maps_leads_krishsax is not in configuration) - resource "forgejo_team_member" "maps_leads_krishsax" { - team_id = 38 -> null - user = "krishsax" -> null } # forgejo_team_member.maps_leads_siddharthr will be created + resource "forgejo_team_member" "maps_leads_siddharthr" { + team_id = 38 + user = "sradhak2" } # forgejo_team_member.maps_leads_xboxbedrock will be destroyed # (because forgejo_team_member.maps_leads_xboxbedrock is not in configuration) - resource "forgejo_team_member" "maps_leads_xboxbedrock" { - team_id = 38 -> null - user = "xboxbedrock" -> null } # forgejo_team_member.occ_acchs will be destroyed # (because forgejo_team_member.occ_acchs is not in configuration) - resource "forgejo_team_member" "occ_acchs" { - team_id = 33 -> null - user = "acchs" -> null } # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place ~ resource "github_repository_ruleset" "cmu-eats_eats_ruleset" { id = "21330218" name = "Default" # (6 unchanged attributes hidden) ~ bypass_actors { ~ actor_id = 0 -> 161775995 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_team_membership.cmu-maps_krishsax will be destroyed # (because github_team_membership.cmu-maps_krishsax is not in configuration) - resource "github_team_membership" "cmu-maps_krishsax" { - etag = "W/\"0ca81c85b299f01891dc704fc1e1f11d9b5e73a022c43d28f0593b116aa3098f\"" -> null - id = "10872360:krishsaxena" -> null - role = "maintainer" -> null - team_id = "10872360" -> null - username = "krishsaxena" -> null } # github_team_membership.cmu-maps_siddharthr will be created + resource "github_team_membership" "cmu-maps_siddharthr" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = "10872360" + username = "overthrowing" } # github_team_membership.cmu-maps_xboxbedrock will be destroyed # (because github_team_membership.cmu-maps_xboxbedrock is not in configuration) - resource "github_team_membership" "cmu-maps_xboxbedrock" { - etag = "W/\"695105377afc546ab133d841be4a870c042dca48a6172b3388875744e9e1aef9\"" -> null - id = "10872360:xboxbedrock" -> null - role = "maintainer" -> null - team_id = "10872360" -> null - username = "xboxbedrock" -> null } # github_team_membership.occ_acchs will be destroyed # (because github_team_membership.occ_acchs is not in configuration) - resource "github_team_membership" "occ_acchs" { - etag = "W/\"886b153b13edb3f78b81ca6ef27b40dc0f1fc0536eeef77c854da6ebd9d7342e\"" -> null - id = "19240310:acchs-at-cmu" -> null - role = "member" -> null - team_id = "19240310" -> null - username = "acchs-at-cmu" -> null } Plan: 3 to add, 1 to change, 7 to destroy. ╷ │ Error: Unable to read team member │ │ User 'krishsax' in team with ID 13 not found: The target couldn't be found. ╵ ``` </details> --- ### Plan Summary 6 projects, 5 with changes, 0 with no changes, 1 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```

Ran Plan for 6 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.21s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # googleplay_app_iam.maps_krishsax will be destroyed
  # (because googleplay_app_iam.maps_krishsax is not in configuration)
- resource "googleplay_app_iam" "maps_krishsax" {
-       app_id               = "com.cmumaps.app" -> null
-       expanded_permissions = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       permissions          = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       user_id              = "krishsax@andrew.cmu.edu" -> null
    }

  # googleplay_app_iam.maps_siddharthr will be created
+ resource "googleplay_app_iam" "maps_siddharthr" {
+       app_id               = "com.cmumaps.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_APP_CONTENT",
+           "CAN_MANAGE_PUBLIC_APKS",
+           "CAN_MANAGE_PUBLIC_LISTING",
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_REPLY_TO_REVIEWS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "sradhak2@andrew.cmu.edu"
    }

  # googleplay_app_iam.maps_xboxbedrock will be destroyed
  # (because googleplay_app_iam.maps_xboxbedrock is not in configuration)
- resource "googleplay_app_iam" "maps_xboxbedrock" {
-       app_id               = "com.cmumaps.app" -> null
-       expanded_permissions = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       permissions          = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       user_id              = "spola2@andrew.cmu.edu" -> null
    }

  # googleplay_user.krishsax will be destroyed
  # (because googleplay_user.krishsax is not in configuration)
- resource "googleplay_user" "krishsax" {
-       email                = "krishsax@andrew.cmu.edu" -> null
-       expanded_permissions = [
-           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ] -> null
-       global_permissions   = [
-           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ] -> null
-       id                   = "krishsax@andrew.cmu.edu" -> null
-       name                 = "developers/7984360456919420393/users/krishsax@andrew.cmu.edu" -> null
    }

  # googleplay_user.siddharthr will be created
+ resource "googleplay_user" "siddharthr" {
+       email                = "sradhak2@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

Plan: 2 to add, 0 to change, 3 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 2 to add, 0 to change, 3 to destroy.


2. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.20s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-maps_ch0_krishsax will be destroyed
  # (because null_resource.slack_cmu-maps_ch0_krishsax is not in configuration)
- resource "null_resource" "slack_cmu-maps_ch0_krishsax" {
-       id       = "8867766601843827684" -> null
-       triggers = {
          - "channel" = "C05T7QRN4LF"
          - "user"    = "U09DL6NH8FL"
        } -> null
    }

  # null_resource.slack_cmu-maps_ch0_siddharthr will be created
+ resource "null_resource" "slack_cmu-maps_ch0_siddharthr" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C05T7QRN4LF"
          + "user"    = "U09DC6MFB5E"
        }
    }

  # null_resource.slack_cmu-maps_ch0_xboxbedrock will be destroyed
  # (because null_resource.slack_cmu-maps_ch0_xboxbedrock is not in configuration)
- resource "null_resource" "slack_cmu-maps_ch0_xboxbedrock" {
-       id       = "2901886759953880290" -> null
-       triggers = {
          - "channel" = "C05T7QRN4LF"
          - "user"    = "U0AQRKYLGHK"
        } -> null
    }

  # null_resource.slack_hub_siddharthr will be created
+ resource "null_resource" "slack_hub_siddharthr" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U09DC6MFB5E"
        }
    }

  # null_resource.slack_leads_siddharthr will be created
+ resource "null_resource" "slack_leads_siddharthr" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0AN1M0HDV0"
          + "user"    = "U09DC6MFB5E"
        }
    }

Plan: 3 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 3 to add, 0 to change, 2 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.15s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.ops_siddharthr will be created
+ resource "google_cloud_identity_group_membership" "ops_siddharthr" {
+       create_time = (known after apply)
+       group       = "groups/03fwokq036zsp1f"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "sradhak2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

  # google_cloud_identity_group_membership.tech_siddharthr will be created
+ resource "google_cloud_identity_group_membership" "tech_siddharthr" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "sradhak2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 2 to add, 0 to change, 0 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-maps_maps will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-maps_maps" {
        id       = "scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64"
!       members  = [
-           "krishsax",
-           "spola2",
+           "sradhak2",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-maps_maps_admins will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-maps_maps_admins" {
        id       = "scottylabs/group-memberships/75db0c7f-7cfb-4d89-b84f-0f50523ed55b"
!       members  = [
-           "krishsax",
-           "spola2",
+           "sradhak2",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "sradhak2",
            # (29 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 3 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 3 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.21s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # discord_member_roles.krishsax will be updated in-place
! resource "discord_member_roles" "krishsax" {
        id        = "1353118209024331948:782966391321395240"
        # (2 unchanged attributes hidden)

-       role {
-           has_role = true -> null
-           role_id  = "1526003694708195448" -> null
        }

        # (3 unchanged blocks hidden)
    }

  # discord_member_roles.siddharthr will be created
+ resource "discord_member_roles" "siddharthr" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "584841698186756167"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1519792320755662928"
        }
+       role {
+           has_role = true
+           role_id  = "1526003694708195448"
        }
    }

  # discord_member_roles.xboxbedrock will be updated in-place
! resource "discord_member_roles" "xboxbedrock" {
        id        = "1353118209024331948:712398210345205871"
        # (2 unchanged attributes hidden)

-       role {
-           has_role = true -> null
-           role_id  = "1526003694708195448" -> null
        }

        # (5 unchanged blocks hidden)
    }

Plan: 1 to add, 2 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 2 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Plan Error

Show Output
running '/nix/store/v8n4x020rdfr57a0s3xdq46a3mdd5bk7-opentofu-1.12.5/bin/tofu plan -input=false -refresh -out /var/lib/atlantis/repos/ScottyLabs/governance/188/default/tofu/git/git-default.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/188/default/tofu/git': exit status 1
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.08s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
github_team.cbp: Refreshing state... [id=17170099]
github_repository.devops_collie: Refreshing state... [id=collie]
github_team_membership.cmu-maps_krishsax: Refreshing state... [id=10872360:krishsaxena]
github_repository.devops_documentation: Refreshing state... [id=documentation]
github_repository.cmu-cal_cal: Refreshing state... [id=cal]
github_repository.cmu-insta_insta: Refreshing state... [id=insta]
github_team_membership.cmu-maps_xboxbedrock: Refreshing state... [id=10872360:xboxbedrock]
github_repository.slai_mcp_server: Refreshing state... [id=mcp-server]
github_repository.lost-and-found_lost_and_found_v2: Refreshing state... [id=lost-and-found-v2]
github_team.cmu-maps: Refreshing state... [id=10872360]
github_repository.devops_infrastructure: Refreshing state... [id=infrastructure]
github_team.cmu-housing: Refreshing state... [id=17937919]
github_repository.cbp_discord_verify: Refreshing state... [id=discord-verify]
github_repository.cmu-courses_internet_archive: Refreshing state... [id=internet-archive]
github_team.cmu-insta: Refreshing state... [id=19240311]
github_repository.cmu-housing_housing: Refreshing state... [id=housing]
github_repository.slai_cmugpt_agent: Refreshing state... [id=cmugpt-agent]
github_team.slai: Refreshing state... [id=17571328]
github_team.uia: Refreshing state... [id=17529763]
github_repository.devops_governance: Refreshing state... [id=governance]
github_repository.cmu-maps_maps: Refreshing state... [id=maps]
github_repository.cbp_bus_sign: Refreshing state... [id=bus-sign]
github_team.terrier: Refreshing state... [id=12028577]
github_repository.cmu-eats_dining_api: Refreshing state... [id=dining-api]
github_repository.cbp_groupme_mirror: Refreshing state... [id=groupme-mirror]
github_repository.tartan-vote_tartan_vote: Refreshing state... [id=tartan-vote]
github_team.quest: Refreshing state... [id=17330197]
github_repository.uia_scottylabs_org: Refreshing state... [id=scottylabs-org]
github_repository.terrier_terrier: Refreshing state... [id=terrier]
github_repository.uia_components: Refreshing state... [id=components]
github_repository.occ_occ: Refreshing state... [id=occ]
github_team.cmu-cal: Refreshing state... [id=15669112]
github_repository.cbp_dalmatian: Refreshing state... [id=dalmatian]
github_team.occ: Refreshing state... [id=19240310]
github_team.lost-and-found: Refreshing state... [id=4088608]
github_repository.slai_cmugpt_sms_surface: Refreshing state... [id=cmugpt-sms-surface]
github_repository.slai_cmugpt_surface: Refreshing state... [id=cmugpt-surface]
github_team.cmu-courses: Refreshing state... [id=3643907]
github_repository.cmu-courses_courses: Refreshing state... [id=courses]
github_repository.cmu-eats_eats: Refreshing state... [id=eats]
github_team.devops: Refreshing state... [id=14706218]
github_repository.devops_kennel: Refreshing state... [id=kennel]
github_repository.quest_quest: Refreshing state... [id=quest]
github_team.cmu-eats: Refreshing state... [id=16807600]
github_repository.devops_link_shortener: Refreshing state... [id=link-shortener]
github_team.tartan-vote: Refreshing state... [id=17701833]
data.external.identity_brightz: Reading...
data.external.identity_brightz: Read complete after 0s [id=-]
data.external.identity_thesuperrl: Reading...
data.external.identity_riztofu: Reading...
data.external.identity_thesuperrl: Read complete after 0s [id=-]
data.external.identity_xboxbedrock: Reading...
data.external.identity_riztofu: Read complete after 0s [id=-]
data.external.identity_gostmeaper: Reading...
data.external.identity_jc_965: Reading...
data.external.identity_xboxbedrock: Read complete after 1s [id=-]
data.external.identity_anish: Reading...
data.external.identity_gostmeaper: Read complete after 1s [id=-]
data.external.identity_kritdass: Reading...
data.external.identity_amzoeee: Reading...
data.external.identity_jc_965: Read complete after 0s [id=-]
data.external.identity_jiyas: Reading...
data.external.identity_anish: Read complete after 0s [id=-]
data.external.identity_emmawong: Reading...
data.external.identity_kritdass: Read complete after 0s [id=-]
data.external.identity_dkou: Reading...
data.external.identity_amzoeee: Read complete after 0s [id=-]
data.external.identity_siddharthr: Reading...
data.external.identity_jiyas: Read complete after 0s [id=-]
data.external.identity_rhsen: Reading...
data.external.identity_emmawong: Read complete after 0s [id=-]
data.external.identity_krishsax: Reading...
data.external.identity_dkou: Read complete after 0s [id=-]
data.external.identity_zoexue: Reading...
data.external.identity_siddharthr: Read complete after 0s [id=-]
data.external.identity_kechezon: Reading...
data.external.identity_rhsen: Read complete after 0s [id=-]
data.external.identity_kothurvidur: Reading...
data.external.identity_krishsax: Read complete after 0s [id=-]
data.external.identity_manrajs: Reading...
data.external.identity_zoexue: Read complete after 0s [id=-]
data.external.identity_etashjha: Reading...
data.external.identity_liamwalk: Reading...
data.external.identity_kothurvidur: Read complete after 0s [id=-]
data.external.identity_tentype: Reading...
data.external.identity_kechezon: Read complete after 1s [id=-]
data.external.identity_nikhilr: Reading...
data.external.identity_manrajs: Read complete after 1s [id=-]
data.external.identity_maybe_yiyi: Reading...
data.external.identity_etashjha: Read complete after 1s [id=-]
data.external.identity_jessica_lr_song: Reading...
data.external.identity_liamwalk: Read complete after 1s [id=-]
data.external.identity_jw8: Reading...
data.external.identity_tentype: Read complete after 1s [id=-]
data.external.identity_acchs: Reading...
data.external.identity_nikhilr: Read complete after 0s [id=-]
data.external.identity_jalenluorion: Reading...
data.external.identity_maybe_yiyi: Read complete after 0s [id=-]
data.external.identity_rzhu3: Reading...
data.external.identity_jessica_lr_song: Read complete after 0s [id=-]
data.external.identity_dcrodrig: Reading...
random_password.kennel_webhook_secret: Refreshing state... [id=none]
forgejo_team_member.cmu-maps_xboxbedrock: Refreshing state...
forgejo_repository.cmu-courses_internet_archive: Refreshing state... [name=internet-archive]
data.external.identity_jw8: Read complete after 0s [id=-]
forgejo_repository.devops_collie: Refreshing state... [name=collie]
data.forgejo_organization.this: Reading...
data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs]
forgejo_team_member.cmu-maps_krishsax: Refreshing state...
forgejo_repository.tartan-vote_tartan_vote: Refreshing state... [name=tartan-vote]
forgejo_repository.cmu-maps_maps: Refreshing state... [name=maps]
data.external.identity_acchs: Read complete after 0s [id=-]
forgejo_repository.cmu-insta_insta: Refreshing state... [name=insta]
forgejo_repository.cmu-eats_dining_api: Refreshing state... [name=dining-api]
forgejo_repository.slai_cmugpt_agent: Refreshing state... [name=cmugpt-agent]
forgejo_repository.quest_quest: Refreshing state... [name=quest]
forgejo_team_member.maps_leads_xboxbedrock: Refreshing state...
forgejo_team_member.maps_leads_krishsax: Refreshing state...
data.external.identity_rzhu3: Read complete after 0s [id=-]
data.external.identity_jalenluorion: Read complete after 0s [id=-]
forgejo_repository.slai_cmugpt_sms_surface: Refreshing state... [name=cmugpt-sms-surface]
forgejo_repository.devops_link_shortener: Refreshing state... [name=link-shortener]
forgejo_repository.uia_components: Refreshing state... [name=components]
forgejo_repository.cbp_discord_verify: Refreshing state... [name=discord-verify]
data.external.identity_dcrodrig: Read complete after 0s [id=-]
forgejo_repository.cbp_bus_sign: Refreshing state... [name=bus-sign]
forgejo_repository.uia_scottylabs_org: Refreshing state... [name=scottylabs-org]
forgejo_repository.cmu-cal_cal: Refreshing state... [name=cal]
forgejo_repository.lost-and-found_lost_and_found_v2: Refreshing state... [name=lost-and-found-v2]
forgejo_repository.cbp_groupme_mirror: Refreshing state... [name=groupme-mirror]
forgejo_repository.devops_kennel: Refreshing state... [name=kennel]
forgejo_repository.terrier_terrier: Refreshing state... [name=terrier]
forgejo_repository.cmu-housing_housing: Refreshing state... [name=housing]
forgejo_repository.devops_infrastructure: Refreshing state... [name=infrastructure]
forgejo_repository.cmu-eats_eats: Refreshing state... [name=eats]
forgejo_repository.cbp_dalmatian: Refreshing state... [name=dalmatian]
forgejo_repository.slai_cmugpt_surface: Refreshing state... [name=cmugpt-surface]
forgejo_repository.devops_governance: Refreshing state... [name=governance]
forgejo_repository.devops_documentation: Refreshing state... [name=documentation]
forgejo_repository.cmu-courses_courses: Refreshing state... [name=courses]
forgejo_repository.occ_occ: Refreshing state... [name=occ]
github_actions_repository_permissions.devops_collie_actions: Refreshing state... [id=collie]
github_actions_repository_permissions.devops_documentation_actions: Refreshing state... [id=documentation]
forgejo_repository.slai_mcp_server: Refreshing state... [name=mcp-server]
github_actions_repository_permissions.lost-and-found_lost_and_found_v2_actions: Refreshing state... [id=lost-and-found-v2]
github_actions_repository_permissions.cmu-cal_cal_actions: Refreshing state... [id=cal]
github_actions_repository_permissions.cbp_discord_verify_actions: Refreshing state... [id=discord-verify]
github_actions_repository_permissions.cmu-insta_insta_actions: Refreshing state... [id=insta]
github_actions_repository_permissions.cmu-courses_internet_archive_actions: Refreshing state... [id=internet-archive]
github_actions_repository_permissions.slai_mcp_server_actions: Refreshing state... [id=mcp-server]
github_actions_repository_permissions.slai_cmugpt_agent_actions: Refreshing state... [id=cmugpt-agent]
github_actions_repository_permissions.devops_infrastructure_actions: Refreshing state... [id=infrastructure]
github_actions_repository_permissions.cmu-housing_housing_actions: Refreshing state... [id=housing]
github_actions_repository_permissions.cbp_groupme_mirror_actions: Refreshing state... [id=groupme-mirror]
github_actions_repository_permissions.devops_governance_actions: Refreshing state... [id=governance]
github_actions_repository_permissions.cmu-maps_maps_actions: Refreshing state... [id=maps]
github_actions_repository_permissions.cbp_bus_sign_actions: Refreshing state... [id=bus-sign]
github_actions_repository_permissions.terrier_terrier_actions: Refreshing state... [id=terrier]
github_actions_repository_permissions.cmu-eats_dining_api_actions: Refreshing state... [id=dining-api]
github_actions_repository_permissions.uia_components_actions: Refreshing state... [id=components]
github_actions_repository_permissions.tartan-vote_tartan_vote_actions: Refreshing state... [id=tartan-vote]
github_actions_repository_permissions.uia_scottylabs_org_actions: Refreshing state... [id=scottylabs-org]
github_actions_repository_permissions.slai_cmugpt_sms_surface_actions: Refreshing state... [id=cmugpt-sms-surface]
github_actions_repository_permissions.slai_cmugpt_surface_actions: Refreshing state... [id=cmugpt-surface]
github_actions_repository_permissions.occ_occ_actions: Refreshing state... [id=occ]
github_team_membership.quest_brightz: Refreshing state... [id=17330197:bright-zheng]
github_actions_repository_permissions.cmu-courses_courses_actions: Refreshing state... [id=courses]
github_team_membership.devops_thesuperrl: Refreshing state... [id=14706218:thesuperrl]
github_team_membership.cmu-housing_xboxbedrock: Refreshing state... [id=17937919:xboxbedrock]
github_team_membership.quest_xboxbedrock: Refreshing state... [id=17330197:xboxbedrock]
github_team_membership.devops_xboxbedrock: Refreshing state... [id=14706218:xboxbedrock]
github_team_membership.cmu-housing_gostmeaper: Refreshing state... [id=17937919:gostmeaper]
github_actions_repository_permissions.cbp_dalmatian_actions: Refreshing state... [id=dalmatian]
github_team_membership.slai_jc_965: Refreshing state... [id=17571328:jc-965]
github_team_membership.devops_anish: Refreshing state... [id=14706218:ap-1]
github_team_membership.cmu-courses_anish: Refreshing state... [id=3643907:ap-1]
github_team_membership.occ_anish: Refreshing state... [id=19240310:ap-1]
github_team_membership.cmu-maps_anish: Refreshing state... [id=10872360:ap-1]
github_team_membership.quest_anish: Refreshing state... [id=17330197:ap-1]
github_team_membership.terrier_anish: Refreshing state... [id=12028577:ap-1]
github_team_membership.terrier_kritdass: Refreshing state... [id=12028577:kritdass]
github_team_membership.cmu-housing_amzoeee: Refreshing state... [id=17937919:amzoeee]
github_team_membership.slai_jiyas: Refreshing state... [id=17571328:jiya2301]
github_team_membership.cmu-housing_emmawong: Refreshing state... [id=17937919:emw868]
github_team_membership.terrier_dkou: Refreshing state... [id=12028577:sharkree]
github_team_membership.cmu-housing_rhsen: Refreshing state... [id=17937919:rohansen08]
github_team_membership.slai_krishsax: Refreshing state... [id=17571328:krishsaxena]
github_team_membership.cmu-housing_zoexue: Refreshing state... [id=17937919:o-zox-o]
github_team_membership.cmu-eats_anish: Refreshing state... [id=16807600:ap-1]
github_team_membership.terrier_kothurvidur: Refreshing state... [id=12028577:kothurvidur]
github_team_membership.quest_kechezon: Refreshing state... [id=17330197:kechezon]
github_team_membership.cmu-housing_manrajs: Refreshing state... [id=17937919:manraj10]
github_team_membership.terrier_etashjha: Refreshing state... [id=12028577:etashj]
github_team_membership.cmu-courses_etashjha: Refreshing state... [id=3643907:etashj]
github_team_membership.cmu-insta_etashjha: Refreshing state... [id=19240311:etashj]
github_team_membership.cbp_liamwalk: Refreshing state... [id=17170099:pandaaured]
github_team_membership.cmu-housing_tentype: Refreshing state... [id=17937919:tentype]
github_team_membership.cbp_tentype: Refreshing state... [id=17170099:tentype]
github_team_membership.cmu-housing_nikhilr: Refreshing state... [id=17937919:ecstaticpilot]
github_team_membership.cbp_maybe_yiyi: Refreshing state... [id=17170099:maybe-yiyi]
github_team_membership.cmu-housing_jessica_lr_song: Refreshing state... [id=17937919:jessica-lr-song]
github_team_membership.tartan-vote_riztofu: Refreshing state... [id=17701833:riztofu]
github_team_membership.tartan-vote_maybe_yiyi: Refreshing state... [id=17701833:maybe-yiyi]
github_team_membership.quest_jw8: Refreshing state... [id=17330197:jeffreyyijiwang]
restapi_object.cmu-courses_internet_archive_push_mirror: Refreshing state... [id=remote_mirror_58X__Yfs0bR]
forgejo_team.groupme-mirror_leads: Refreshing state... [name=groupme-mirror-leads]
forgejo_team.cmugpt-sms-surface_leads: Refreshing state... [name=cmugpt-sms-surface-leads]
forgejo_team.cmugpt-agent_leads: Refreshing state... [name=cmugpt-agent-leads]
forgejo_team.insta_leads: Refreshing state... [name=insta-leads]
forgejo_team.lost-and-found: Refreshing state... [name=lost-and-found]
forgejo_team.quest: Refreshing state... [name=quest]
forgejo_team.tartan-vote_leads: Refreshing state... [name=tartan-vote-leads]
forgejo_team.quest_leads: Refreshing state... [name=quest-leads]
forgejo_team.devops_leads: Refreshing state... [name=devops-leads]
forgejo_team.cmugpt-surface_leads: Refreshing state... [name=cmugpt-surface-leads]
forgejo_team.eats_leads: Refreshing state... [name=eats-leads]
forgejo_team.link-shortener_leads: Refreshing state... [name=link-shortener-leads]
forgejo_team.documentation_leads: Refreshing state... [name=documentation-leads]
forgejo_team.cmugpt-mcp-server_leads: Refreshing state... [name=cmugpt-mcp-server-leads]
forgejo_team.cmu-courses: Refreshing state... [name=cmu-courses]
forgejo_team.cmu-insta: Refreshing state... [name=cmu-insta]
forgejo_team.cmu-housing: Refreshing state... [name=cmu-housing]
forgejo_team.cal_leads: Refreshing state... [name=cal-leads]
forgejo_team.terrier_leads: Refreshing state... [name=terrier-leads]
forgejo_team.cmu-maps: Refreshing state... [name=cmu-maps]
forgejo_team.housing_leads: Refreshing state... [name=housing-leads]
forgejo_team.maps_leads: Refreshing state... [name=maps-leads]
forgejo_team.bus-sign_leads: Refreshing state... [name=bus-sign-leads]
forgejo_team.cmu-courses_leads: Refreshing state... [name=cmu-courses-leads]
forgejo_team.uia: Refreshing state... [name=uia]
forgejo_team.terrier: Refreshing state... [name=terrier]
forgejo_team.occ: Refreshing state... [name=occ]
forgejo_team.devops: Refreshing state... [name=devops]
forgejo_team.cbp: Refreshing state... [name=cbp]
forgejo_team.cmu-cal: Refreshing state... [name=cmu-cal]
forgejo_team.slai: Refreshing state... [name=slai]
forgejo_team.occ-main_leads: Refreshing state... [name=occ-main-leads]
forgejo_team.courses_leads: Refreshing state... [name=courses-leads]
forgejo_team.tartan-vote: Refreshing state... [name=tartan-vote]
forgejo_team.discord-verify_leads: Refreshing state... [name=discord-verify-leads]
forgejo_team.cmu-eats: Refreshing state... [name=cmu-eats]
forgejo_team.dalmatian_leads: Refreshing state... [name=dalmatian-leads]
restapi_object.devops_collie_push_mirror: Refreshing state... [id=remote_mirror_ljvVejNFXaS]
github_team_membership.occ_acchs: Refreshing state... [id=19240310:acchs-at-cmu]
restapi_object.tartan-vote_tartan_vote_push_mirror: Refreshing state... [id=remote_mirror_EUFq4sFSnH5]
restapi_object.tartan-vote_tartan_vote_docs_webhook: Refreshing state... [id=25]
restapi_object.cmu-maps_maps_push_mirror: Refreshing state... [id=remote_mirror_e4_FUnvoslJ]
restapi_object.cmu-eats_dining_api_push_mirror: Refreshing state... [id=remote_mirror_a4MsQ5rO1_Z]
restapi_object.cmu-insta_insta_push_mirror: Refreshing state... [id=remote_mirror_891nu6XQ1us]
github_actions_repository_permissions.cmu-eats_eats_actions: Refreshing state... [id=eats]
github_team_membership.slai_rzhu3: Refreshing state... [id=17571328:richard-31415]
github_team_membership.cmu-cal_jalenluorion: Refreshing state... [id=15669112:jalenluorion]
restapi_object.slai_cmugpt_agent_push_mirror: Refreshing state... [id=remote_mirror_kNY6rVcxdyr]
github_team_membership.occ_dcrodrig: Refreshing state... [id=19240310:newhacker1746]
restapi_object.slai_cmugpt_sms_surface_push_mirror: Refreshing state... [id=remote_mirror__fkBkASN2jA]
restapi_object.quest_quest_push_mirror: Refreshing state... [id=remote_mirror_U9d9BPAon29]
restapi_object.cbp_discord_verify_push_mirror: Refreshing state... [id=remote_mirror_3MV7Up6cY1X]
restapi_object.cbp_bus_sign_push_mirror: Refreshing state... [id=remote_mirror_MkKxQ2vA9pc]
restapi_object.devops_link_shortener_push_mirror: Refreshing state... [id=remote_mirror_cp8daMpt82v]
restapi_object.uia_components_push_mirror: Refreshing state... [id=remote_mirror_Kwwa8TDnTgc]
restapi_object.uia_scottylabs_org_push_mirror: Refreshing state... [id=remote_mirror_h6yN6QdS1lD]
github_actions_repository_permissions.devops_kennel_actions: Refreshing state... [id=kennel]
restapi_object.lost-and-found_lost_and_found_v2_push_mirror: Refreshing state... [id=remote_mirror_n_UZBFU-CY4]
restapi_object.cmu-cal_cal_push_mirror: Refreshing state... [id=remote_mirror_CO243uIONoq]
restapi_object.terrier_terrier_push_mirror: Refreshing state... [id=remote_mirror_YopDgZksZv5]
restapi_object.cmu-housing_housing_push_mirror: Refreshing state... [id=remote_mirror_uoV2HG4aiOT]
restapi_object.cmu-housing_housing_docs_webhook: Refreshing state... [id=15]
restapi_object.devops_infrastructure_push_mirror: Refreshing state... [id=remote_mirror_oYtFpKVzTaZ]
restapi_object.cbp_groupme_mirror_push_mirror: Refreshing state... [id=remote_mirror_U61HJHB9VGE]
restapi_object.devops_kennel_push_mirror: Refreshing state... [id=remote_mirror_cT9psJkhTQ4]
restapi_object.cmu-eats_eats_push_mirror: Refreshing state... [id=remote_mirror_sjYwMA6NPgh]
restapi_object.cbp_dalmatian_push_mirror: Refreshing state... [id=remote_mirror_qzlCpLK153F]
restapi_object.cbp_dalmatian_docs_webhook: Refreshing state... [id=7]
restapi_object.slai_cmugpt_surface_push_mirror: Refreshing state... [id=remote_mirror_cPTQN6S5SpB]
restapi_object.devops_documentation_push_mirror: Refreshing state... [id=remote_mirror_izmsqaNB6Wx]
restapi_object.devops_governance_push_mirror: Refreshing state... [id=remote_mirror_lt9JzkX1jMq]
restapi_object.cmu-courses_courses_push_mirror: Refreshing state... [id=remote_mirror_WMR9yQwKFDF]
restapi_object.slai_mcp_server_push_mirror: Refreshing state... [id=remote_mirror_Tqh2y0R_vDp]
restapi_object.occ_occ_push_mirror: Refreshing state... [id=remote_mirror_moBLq1YG1qz]
github_actions_repository_permissions.quest_quest_actions: Refreshing state... [id=quest]
github_actions_repository_permissions.devops_link_shortener_actions: Refreshing state... [id=link-shortener]
restapi_object.uia_scottylabs_org_kennel_webhook: Refreshing state... [id=29]
restapi_object.devops_kennel_kennel_webhook: Refreshing state... [id=14]
restapi_object.devops_infrastructure_kennel_webhook: Refreshing state... [id=26]
restapi_object.slai_cmugpt_surface_kennel_webhook: Refreshing state... [id=23]
restapi_object.slai_cmugpt_sms_surface_kennel_webhook: Refreshing state... [id=19]
restapi_object.tartan-vote_tartan_vote_kennel_webhook: Refreshing state... [id=24]
restapi_object.cmu-housing_housing_kennel_webhook: Refreshing state... [id=16]
restapi_object.cmu-courses_courses_kennel_webhook: Refreshing state... [id=11]
restapi_object.devops_collie_kennel_webhook: Refreshing state... [id=17]
restapi_object.cbp_dalmatian_kennel_webhook: Refreshing state... [id=6]
restapi_object.slai_cmugpt_agent_kennel_webhook: Refreshing state... [id=28]
restapi_object.cmu-eats_eats_kennel_webhook: Refreshing state... [id=31]
restapi_object.quest_quest_kennel_webhook: Refreshing state... [id=10]
restapi_object.devops_link_shortener_kennel_webhook: Refreshing state... [id=8]
restapi_object.cbp_discord_verify_kennel_webhook: Refreshing state... [id=18]
restapi_object.cmu-cal_cal_kennel_webhook: Refreshing state... [id=20]
restapi_object.terrier_terrier_kennel_webhook: Refreshing state... [id=12]
restapi_object.slai_mcp_server_kennel_webhook: Refreshing state... [id=13]
restapi_object.devops_governance_kennel_webhook: Refreshing state... [id=22]
restapi_object.cbp_bus_sign_kennel_webhook: Refreshing state... [id=21]
restapi_object.uia_components_kennel_webhook: Refreshing state... [id=27]
restapi_object.cmu-maps_maps_kennel_webhook: Refreshing state... [id=9]
github_repository_deploy_key.cmu-courses_internet_archive_mirror_deploy_key: Refreshing state... [id=internet-archive:159665356]
forgejo_team_member.groupme-mirror_leads_tentype: Refreshing state...
forgejo_team_member.groupme-mirror_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.cmugpt-sms-surface_leads_krishsax: Refreshing state...
forgejo_team_member.cmugpt-agent_leads_krishsax: Refreshing state...
forgejo_team_member.insta_leads_etashjha: Refreshing state...
forgejo_team_repository.lost-and-found_lost_and_found_v2: Refreshing state...
forgejo_team_member.quest_brightz: Refreshing state...
forgejo_team_member.quest_jw8: Refreshing state...
forgejo_team_repository.quest_quest: Refreshing state...
forgejo_team_member.quest_xboxbedrock: Refreshing state...
forgejo_team_member.quest_anish: Refreshing state...
forgejo_team_member.quest_kechezon: Refreshing state...
forgejo_team_member.tartan-vote_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.quest_leads_xboxbedrock: Refreshing state...
forgejo_team_member.quest_leads_anish: Refreshing state...
forgejo_team_member.devops_leads_thesuperrl: Refreshing state...
forgejo_team_member.devops_leads_anish: Refreshing state...
forgejo_team_member.cmugpt-surface_leads_krishsax: Refreshing state...
forgejo_team_member.eats_leads_anish: Refreshing state...
forgejo_team_member.link-shortener_leads_thesuperrl: Refreshing state...
forgejo_team_member.link-shortener_leads_anish: Refreshing state...
forgejo_team_member.documentation_leads_thesuperrl: Refreshing state...
forgejo_team_member.documentation_leads_anish: Refreshing state...
forgejo_team_member.cmugpt-mcp-server_leads_krishsax: Refreshing state...
forgejo_team_member.cmu-courses_anish: Refreshing state...
forgejo_team_repository.cmu-courses_courses: Refreshing state...
forgejo_team_member.cmu-courses_etashjha: Refreshing state...
forgejo_team_repository.cmu-courses_internet_archive: Refreshing state...
forgejo_team_repository.cmu-insta_insta: Refreshing state...
forgejo_team_member.cmu-insta_etashjha: Refreshing state...
forgejo_team_member.cmu-housing_emmawong: Refreshing state...
forgejo_team_member.cmu-housing_xboxbedrock: Refreshing state...
forgejo_team_member.cmu-housing_gostmeaper: Refreshing state...
forgejo_team_member.cmu-housing_manrajs: Refreshing state...
forgejo_team_member.cmu-housing_nikhilr: Refreshing state...
forgejo_team_member.cmu-housing_jessica_lr_song: Refreshing state...
forgejo_team_member.cmu-housing_rhsen: Refreshing state...
forgejo_team_member.cmu-housing_zoexue: Refreshing state...
forgejo_team_member.cmu-housing_amzoeee: Refreshing state...
forgejo_team_repository.cmu-housing_housing: Refreshing state...
forgejo_team_member.cmu-housing_tentype: Refreshing state...
forgejo_team_member.cal_leads_jalenluorion: Refreshing state...
forgejo_team_member.terrier_leads_kritdass: Refreshing state...
forgejo_team_member.terrier_leads_anish: Refreshing state...
forgejo_team_repository.cmu-maps_maps: Refreshing state...
forgejo_team_member.cmu-maps_anish: Refreshing state...
forgejo_team_member.housing_leads_emmawong: Refreshing state...
forgejo_team_member.housing_leads_gostmeaper: Refreshing state...
forgejo_team_member.housing_leads_nikhilr: Refreshing state...
forgejo_team_member.maps_leads_anish: Refreshing state...
forgejo_team_member.housing_leads_tentype: Refreshing state...
forgejo_team_member.bus-sign_leads_tentype: Refreshing state...
forgejo_team_member.bus-sign_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.cmu-courses_leads_etashjha: Refreshing state...
forgejo_team_member.cmu-courses_leads_anish: Refreshing state...
forgejo_team_repository.uia_components: Refreshing state...
forgejo_team_repository.uia_scottylabs_org: Refreshing state...
forgejo_team_member.devops_anish: Refreshing state...
forgejo_team_member.devops_xboxbedrock: Refreshing state...
forgejo_team_repository.devops_link_shortener: Refreshing state...
forgejo_team_repository.devops_governance: Refreshing state...
forgejo_team_repository.devops_collie: Refreshing state...
forgejo_team_repository.devops_documentation: Refreshing state...
forgejo_team_repository.devops_kennel: Refreshing state...
forgejo_team_repository.devops_infrastructure: Refreshing state...
forgejo_team_member.devops_thesuperrl: Refreshing state...
forgejo_team_member.terrier_etashjha: Refreshing state...
forgejo_team_member.terrier_anish: Refreshing state...
forgejo_team_repository.terrier_terrier: Refreshing state...
forgejo_team_member.terrier_kritdass: Refreshing state...
forgejo_team_member.terrier_dkou: Refreshing state...
forgejo_team_member.terrier_kothurvidur: Refreshing state...
forgejo_team_member.occ_dcrodrig: Refreshing state...
forgejo_team_member.occ_acchs: Refreshing state...
forgejo_team_repository.occ_occ: Refreshing state...
forgejo_team_member.occ_anish: Refreshing state...
forgejo_team_repository.cbp_discord_verify: Refreshing state...
forgejo_team_member.cbp_liamwalk: Refreshing state...
forgejo_team_member.cbp_maybe_yiyi: Refreshing state...
forgejo_team_repository.cbp_bus_sign: Refreshing state...
forgejo_team_repository.cbp_groupme_mirror: Refreshing state...
forgejo_team_member.cbp_tentype: Refreshing state...
forgejo_team_repository.cbp_dalmatian: Refreshing state...
forgejo_team_repository.cmu-cal_cal: Refreshing state...
forgejo_team_member.cmu-cal_jalenluorion: Refreshing state...
forgejo_team_repository.slai_cmugpt_sms_surface: Refreshing state...
forgejo_team_repository.slai_cmugpt_surface: Refreshing state...
forgejo_team_repository.slai_cmugpt_agent: Refreshing state...
forgejo_team_repository.slai_mcp_server: Refreshing state...
forgejo_team_member.slai_jc_965: Refreshing state...
forgejo_team_member.slai_jiyas: Refreshing state...
forgejo_team_member.slai_rzhu3: Refreshing state...
forgejo_team_member.slai_krishsax: Refreshing state...
forgejo_team_member.occ-main_leads_dcrodrig: Refreshing state...
forgejo_team_member.occ-main_leads_anish: Refreshing state...
forgejo_team_member.courses_leads_etashjha: Refreshing state...
forgejo_team_member.courses_leads_anish: Refreshing state...
forgejo_team_member.tartan-vote_maybe_yiyi: Refreshing state...
forgejo_team_repository.tartan-vote_tartan_vote: Refreshing state...
forgejo_team_member.tartan-vote_riztofu: Refreshing state...
forgejo_team_member.discord-verify_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.discord-verify_leads_tentype: Refreshing state...
forgejo_team_repository.cmu-eats_eats: Refreshing state...
forgejo_team_repository.cmu-eats_dining_api: Refreshing state...
forgejo_team_member.cmu-eats_anish: Refreshing state...
forgejo_team_member.dalmatian_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.dalmatian_leads_tentype: Refreshing state...
github_repository_deploy_key.devops_collie_mirror_deploy_key: Refreshing state... [id=collie:159665332]
github_repository_deploy_key.tartan-vote_tartan_vote_mirror_deploy_key: Refreshing state... [id=tartan-vote:159665350]
github_repository_deploy_key.cmu-maps_maps_mirror_deploy_key: Refreshing state... [id=maps:159665325]
github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Refreshing state... [id=dining-api:161772815]
github_repository_deploy_key.cmu-insta_insta_mirror_deploy_key: Refreshing state... [id=insta:161720245]
github_repository_deploy_key.slai_cmugpt_sms_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-sms-surface:159665365]
github_repository_deploy_key.slai_cmugpt_agent_mirror_deploy_key: Refreshing state... [id=cmugpt-agent:159665360]
github_repository_deploy_key.quest_quest_mirror_deploy_key: Refreshing state... [id=quest:159665323]
github_repository_deploy_key.cbp_discord_verify_mirror_deploy_key: Refreshing state... [id=discord-verify:159665340]
github_repository_deploy_key.cbp_bus_sign_mirror_deploy_key: Refreshing state... [id=bus-sign:159665342]
github_repository_deploy_key.devops_link_shortener_mirror_deploy_key: Refreshing state... [id=link-shortener:159665345]
github_repository_deploy_key.uia_components_mirror_deploy_key: Refreshing state... [id=components:159665353]
github_repository_deploy_key.uia_scottylabs_org_mirror_deploy_key: Refreshing state... [id=scottylabs-org:159665361]
github_repository_deploy_key.lost-and-found_lost_and_found_v2_mirror_deploy_key: Refreshing state... [id=lost-and-found-v2:159665337]
github_repository_deploy_key.cmu-cal_cal_mirror_deploy_key: Refreshing state... [id=cal:159665344]
github_repository_deploy_key.terrier_terrier_mirror_deploy_key: Refreshing state... [id=terrier:159665339]
github_repository_deploy_key.cmu-housing_housing_mirror_deploy_key: Refreshing state... [id=housing:159665369]
github_repository_deploy_key.devops_infrastructure_mirror_deploy_key: Refreshing state... [id=infrastructure:159665343]
github_repository_deploy_key.cbp_groupme_mirror_mirror_deploy_key: Refreshing state... [id=groupme-mirror:159665366]
github_repository_deploy_key.devops_kennel_mirror_deploy_key: Refreshing state... [id=kennel:159665328]
github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Refreshing state... [id=eats:161775995]
github_repository_deploy_key.cbp_dalmatian_mirror_deploy_key: Refreshing state... [id=dalmatian:159665367]
github_repository_deploy_key.slai_cmugpt_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-surface:159665348]
github_repository_deploy_key.devops_documentation_mirror_deploy_key: Refreshing state... [id=documentation:159665362]
github_repository_deploy_key.devops_governance_mirror_deploy_key: Refreshing state... [id=governance:159665346]
github_repository_deploy_key.cmu-courses_courses_mirror_deploy_key: Refreshing state... [id=courses:159665327]
github_repository_deploy_key.slai_mcp_server_mirror_deploy_key: Refreshing state... [id=mcp-server:159665358]
github_repository_deploy_key.occ_occ_mirror_deploy_key: Refreshing state... [id=occ:161720242]
forgejo_team_repository.quest_leads_quest: Refreshing state...
forgejo_team_repository.courses_leads_courses: Refreshing state...
forgejo_team_repository.cmu-courses_leads_internet_archive: Refreshing state...
forgejo_team_repository.insta_leads_insta: Refreshing state...
forgejo_team_repository.housing_leads_housing: Refreshing state...
forgejo_team_repository.maps_leads_maps: Refreshing state...
forgejo_team_repository.devops_leads_governance: Refreshing state...
forgejo_team_repository.devops_leads_infrastructure: Refreshing state...
forgejo_team_repository.devops_leads_collie: Refreshing state...
forgejo_team_repository.documentation_leads_documentation: Refreshing state...
forgejo_team_repository.devops_leads_kennel: Refreshing state...
forgejo_team_repository.link-shortener_leads_link_shortener: Refreshing state...
forgejo_team_repository.terrier_leads_terrier: Refreshing state...
forgejo_team_repository.occ-main_leads_occ: Refreshing state...
forgejo_team_repository.cal_leads_cal: Refreshing state...
forgejo_team_repository.discord-verify_leads_discord_verify: Refreshing state...
forgejo_team_repository.bus-sign_leads_bus_sign: Refreshing state...
forgejo_team_repository.dalmatian_leads_dalmatian: Refreshing state...
forgejo_team_repository.cmugpt-sms-surface_leads_cmugpt_sms_surface: Refreshing state...
forgejo_team_repository.cmugpt-mcp-server_leads_mcp_server: Refreshing state...
forgejo_team_repository.groupme-mirror_leads_groupme_mirror: Refreshing state...
forgejo_team_repository.cmugpt-agent_leads_cmugpt_agent: Refreshing state...
forgejo_team_repository.cmugpt-surface_leads_cmugpt_surface: Refreshing state...
forgejo_team_repository.tartan-vote_leads_tartan_vote: Refreshing state...
forgejo_team_repository.eats_leads_eats: Refreshing state...
forgejo_team_repository.eats_leads_dining_api: Refreshing state...
github_repository_ruleset.cmu-courses_internet_archive_ruleset: Refreshing state... [id=11988074]
github_repository_ruleset.devops_collie_ruleset: Refreshing state... [id=18865204]
github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Refreshing state... [id=16848966]
github_repository_ruleset.cmu-maps_maps_ruleset: Refreshing state... [id=18864961]
github_repository_ruleset.cmu-eats_dining_api_ruleset: Refreshing state... [id=21879626]
github_repository_ruleset.cmu-insta_insta_ruleset: Refreshing state... [id=21841636]
github_repository_ruleset.slai_cmugpt_agent_ruleset: Refreshing state... [id=16476661]
github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Refreshing state... [id=17653622]
github_repository_ruleset.quest_quest_ruleset: Refreshing state... [id=15750052]
github_repository_ruleset.cbp_discord_verify_ruleset: Refreshing state... [id=11449154]
github_repository_ruleset.cbp_bus_sign_ruleset: Refreshing state... [id=15236550]
github_repository_ruleset.devops_link_shortener_ruleset: Refreshing state... [id=17911647]
github_repository_ruleset.uia_components_ruleset: Refreshing state... [id=16366085]
github_repository_ruleset.uia_scottylabs_org_ruleset: Refreshing state... [id=18112214]
github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Refreshing state... [id=18920618]
github_repository_ruleset.cmu-cal_cal_ruleset: Refreshing state... [id=18919281]
github_repository_ruleset.terrier_terrier_ruleset: Refreshing state... [id=12671003]
github_repository_ruleset.cmu-housing_housing_ruleset: Refreshing state... [id=17421604]
github_repository_ruleset.devops_infrastructure_ruleset: Refreshing state... [id=11275118]
github_repository_ruleset.cbp_groupme_mirror_ruleset: Refreshing state... [id=12503296]
github_repository_ruleset.devops_kennel_ruleset: Refreshing state... [id=13193453]
github_repository_ruleset.cmu-eats_eats_ruleset: Refreshing state... [id=21330218]
github_repository_ruleset.cbp_dalmatian_ruleset: Refreshing state... [id=17238915]
github_repository_ruleset.slai_cmugpt_surface_ruleset: Refreshing state... [id=16476662]
github_repository_ruleset.devops_documentation_ruleset: Refreshing state... [id=17079368]
github_repository_ruleset.devops_governance_ruleset: Refreshing state... [id=14488792]
github_repository_ruleset.cmu-courses_courses_ruleset: Refreshing state... [id=15855161]
github_repository_ruleset.slai_mcp_server_ruleset: Refreshing state... [id=17082154]
github_repository_ruleset.occ_occ_ruleset: Refreshing state... [id=21841634]

OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
  + create
  ~ update in-place (current -> planned)
  - destroy

OpenTofu planned the following actions, but then encountered a problem:

  # forgejo_team_member.cmu-maps_siddharthr will be created
  + resource "forgejo_team_member" "cmu-maps_siddharthr" {
      + team_id = 13
      + user    = "sradhak2"
    }

  # forgejo_team_member.cmu-maps_xboxbedrock will be destroyed
  # (because forgejo_team_member.cmu-maps_xboxbedrock is not in configuration)
  - resource "forgejo_team_member" "cmu-maps_xboxbedrock" {
      - team_id = 13 -> null
      - user    = "xboxbedrock" -> null
    }

  # forgejo_team_member.maps_leads_krishsax will be destroyed
  # (because forgejo_team_member.maps_leads_krishsax is not in configuration)
  - resource "forgejo_team_member" "maps_leads_krishsax" {
      - team_id = 38 -> null
      - user    = "krishsax" -> null
    }

  # forgejo_team_member.maps_leads_siddharthr will be created
  + resource "forgejo_team_member" "maps_leads_siddharthr" {
      + team_id = 38
      + user    = "sradhak2"
    }

  # forgejo_team_member.maps_leads_xboxbedrock will be destroyed
  # (because forgejo_team_member.maps_leads_xboxbedrock is not in configuration)
  - resource "forgejo_team_member" "maps_leads_xboxbedrock" {
      - team_id = 38 -> null
      - user    = "xboxbedrock" -> null
    }

  # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place
  ~ resource "github_repository_ruleset" "cmu-eats_eats_ruleset" {
        id          = "21330218"
        name        = "Default"
        # (6 unchanged attributes hidden)

      ~ bypass_actors {
          ~ actor_id    = 0 -> 161775995
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_team_membership.cmu-maps_krishsax will be destroyed
  # (because github_team_membership.cmu-maps_krishsax is not in configuration)
  - resource "github_team_membership" "cmu-maps_krishsax" {
      - etag     = "W/\"0ca81c85b299f01891dc704fc1e1f11d9b5e73a022c43d28f0593b116aa3098f\"" -> null
      - id       = "10872360:krishsaxena" -> null
      - role     = "maintainer" -> null
      - team_id  = "10872360" -> null
      - username = "krishsaxena" -> null
    }

  # github_team_membership.cmu-maps_siddharthr will be created
  + resource "github_team_membership" "cmu-maps_siddharthr" {
      + etag     = (known after apply)
      + id       = (known after apply)
      + role     = "maintainer"
      + team_id  = "10872360"
      + username = "overthrowing"
    }

  # github_team_membership.cmu-maps_xboxbedrock will be destroyed
  # (because github_team_membership.cmu-maps_xboxbedrock is not in configuration)
  - resource "github_team_membership" "cmu-maps_xboxbedrock" {
      - etag     = "W/\"695105377afc546ab133d841be4a870c042dca48a6172b3388875744e9e1aef9\"" -> null
      - id       = "10872360:xboxbedrock" -> null
      - role     = "maintainer" -> null
      - team_id  = "10872360" -> null
      - username = "xboxbedrock" -> null
    }

Plan: 3 to add, 1 to change, 5 to destroy.
╷
│ Error: Unable to read team member
│ 
│ User 'krishsax' in team with ID 13 not found: The target couldn't be found.
╵



Plan Summary

6 projects, 5 with changes, 0 with no changes, 1 failed

  • To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.21s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # googleplay_app_iam.maps_krishsax will be destroyed # (because googleplay_app_iam.maps_krishsax is not in configuration) - resource "googleplay_app_iam" "maps_krishsax" { - app_id = "com.cmumaps.app" -> null - expanded_permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - user_id = "krishsax@andrew.cmu.edu" -> null } # googleplay_app_iam.maps_siddharthr will be created + resource "googleplay_app_iam" "maps_siddharthr" { + app_id = "com.cmumaps.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_APP_CONTENT", + "CAN_MANAGE_PUBLIC_APKS", + "CAN_MANAGE_PUBLIC_LISTING", + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_REPLY_TO_REVIEWS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "sradhak2@andrew.cmu.edu" } # googleplay_app_iam.maps_xboxbedrock will be destroyed # (because googleplay_app_iam.maps_xboxbedrock is not in configuration) - resource "googleplay_app_iam" "maps_xboxbedrock" { - app_id = "com.cmumaps.app" -> null - expanded_permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - user_id = "spola2@andrew.cmu.edu" -> null } # googleplay_user.krishsax will be destroyed # (because googleplay_user.krishsax is not in configuration) - resource "googleplay_user" "krishsax" { - email = "krishsax@andrew.cmu.edu" -> null - expanded_permissions = [ - "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] -> null - global_permissions = [ - "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] -> null - id = "krishsax@andrew.cmu.edu" -> null - name = "developers/7984360456919420393/users/krishsax@andrew.cmu.edu" -> null } # googleplay_user.siddharthr will be created + resource "googleplay_user" "siddharthr" { + email = "sradhak2@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } Plan: 2 to add, 0 to change, 3 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 2 to add, 0 to change, 3 to destroy. --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.20s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # null_resource.slack_cmu-maps_ch0_krishsax will be destroyed # (because null_resource.slack_cmu-maps_ch0_krishsax is not in configuration) - resource "null_resource" "slack_cmu-maps_ch0_krishsax" { - id = "8867766601843827684" -> null - triggers = { - "channel" = "C05T7QRN4LF" - "user" = "U09DL6NH8FL" } -> null } # null_resource.slack_cmu-maps_ch0_siddharthr will be created + resource "null_resource" "slack_cmu-maps_ch0_siddharthr" { + id = (known after apply) + triggers = { + "channel" = "C05T7QRN4LF" + "user" = "U09DC6MFB5E" } } # null_resource.slack_cmu-maps_ch0_xboxbedrock will be destroyed # (because null_resource.slack_cmu-maps_ch0_xboxbedrock is not in configuration) - resource "null_resource" "slack_cmu-maps_ch0_xboxbedrock" { - id = "2901886759953880290" -> null - triggers = { - "channel" = "C05T7QRN4LF" - "user" = "U0AQRKYLGHK" } -> null } # null_resource.slack_hub_siddharthr will be created + resource "null_resource" "slack_hub_siddharthr" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U09DC6MFB5E" } } # null_resource.slack_leads_siddharthr will be created + resource "null_resource" "slack_leads_siddharthr" { + id = (known after apply) + triggers = { + "channel" = "C0AN1M0HDV0" + "user" = "U09DC6MFB5E" } } Plan: 3 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 3 to add, 0 to change, 2 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.15s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.ops_siddharthr will be created + resource "google_cloud_identity_group_membership" "ops_siddharthr" { + create_time = (known after apply) + group = "groups/03fwokq036zsp1f" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "sradhak2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } # google_cloud_identity_group_membership.tech_siddharthr will be created + resource "google_cloud_identity_group_membership" "tech_siddharthr" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "sradhak2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-maps_maps will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-maps_maps" { id = "scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64" ! members = [ - "krishsax", - "spola2", + "sradhak2", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-maps_maps_admins will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-maps_maps_admins" { id = "scottylabs/group-memberships/75db0c7f-7cfb-4d89-b84f-0f50523ed55b" ! members = [ - "krishsax", - "spola2", + "sradhak2", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "sradhak2", # (29 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 3 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 3 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.21s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # discord_member_roles.krishsax will be updated in-place ! resource "discord_member_roles" "krishsax" { id = "1353118209024331948:782966391321395240" # (2 unchanged attributes hidden) - role { - has_role = true -> null - role_id = "1526003694708195448" -> null } # (3 unchanged blocks hidden) } # discord_member_roles.siddharthr will be created + resource "discord_member_roles" "siddharthr" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "584841698186756167" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1519792320755662928" } + role { + has_role = true + role_id = "1526003694708195448" } } # discord_member_roles.xboxbedrock will be updated in-place ! resource "discord_member_roles" "xboxbedrock" { id = "1353118209024331948:712398210345205871" # (2 unchanged attributes hidden) - role { - has_role = true -> null - role_id = "1526003694708195448" -> null } # (5 unchanged blocks hidden) } Plan: 1 to add, 2 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 2 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` **Plan Error** <details><summary>Show Output</summary> ``` running '/nix/store/v8n4x020rdfr57a0s3xdq46a3mdd5bk7-opentofu-1.12.5/bin/tofu plan -input=false -refresh -out /var/lib/atlantis/repos/ScottyLabs/governance/188/default/tofu/git/git-default.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/188/default/tofu/git': exit status 1 Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.08s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ github_team.cbp: Refreshing state... [id=17170099] github_repository.devops_collie: Refreshing state... [id=collie] github_team_membership.cmu-maps_krishsax: Refreshing state... [id=10872360:krishsaxena] github_repository.devops_documentation: Refreshing state... [id=documentation] github_repository.cmu-cal_cal: Refreshing state... [id=cal] github_repository.cmu-insta_insta: Refreshing state... [id=insta] github_team_membership.cmu-maps_xboxbedrock: Refreshing state... [id=10872360:xboxbedrock] github_repository.slai_mcp_server: Refreshing state... [id=mcp-server] github_repository.lost-and-found_lost_and_found_v2: Refreshing state... [id=lost-and-found-v2] github_team.cmu-maps: Refreshing state... [id=10872360] github_repository.devops_infrastructure: Refreshing state... [id=infrastructure] github_team.cmu-housing: Refreshing state... [id=17937919] github_repository.cbp_discord_verify: Refreshing state... [id=discord-verify] github_repository.cmu-courses_internet_archive: Refreshing state... [id=internet-archive] github_team.cmu-insta: Refreshing state... [id=19240311] github_repository.cmu-housing_housing: Refreshing state... [id=housing] github_repository.slai_cmugpt_agent: Refreshing state... [id=cmugpt-agent] github_team.slai: Refreshing state... [id=17571328] github_team.uia: Refreshing state... [id=17529763] github_repository.devops_governance: Refreshing state... [id=governance] github_repository.cmu-maps_maps: Refreshing state... [id=maps] github_repository.cbp_bus_sign: Refreshing state... [id=bus-sign] github_team.terrier: Refreshing state... [id=12028577] github_repository.cmu-eats_dining_api: Refreshing state... [id=dining-api] github_repository.cbp_groupme_mirror: Refreshing state... [id=groupme-mirror] github_repository.tartan-vote_tartan_vote: Refreshing state... [id=tartan-vote] github_team.quest: Refreshing state... [id=17330197] github_repository.uia_scottylabs_org: Refreshing state... [id=scottylabs-org] github_repository.terrier_terrier: Refreshing state... [id=terrier] github_repository.uia_components: Refreshing state... [id=components] github_repository.occ_occ: Refreshing state... [id=occ] github_team.cmu-cal: Refreshing state... [id=15669112] github_repository.cbp_dalmatian: Refreshing state... [id=dalmatian] github_team.occ: Refreshing state... [id=19240310] github_team.lost-and-found: Refreshing state... [id=4088608] github_repository.slai_cmugpt_sms_surface: Refreshing state... [id=cmugpt-sms-surface] github_repository.slai_cmugpt_surface: Refreshing state... [id=cmugpt-surface] github_team.cmu-courses: Refreshing state... [id=3643907] github_repository.cmu-courses_courses: Refreshing state... [id=courses] github_repository.cmu-eats_eats: Refreshing state... [id=eats] github_team.devops: Refreshing state... [id=14706218] github_repository.devops_kennel: Refreshing state... [id=kennel] github_repository.quest_quest: Refreshing state... [id=quest] github_team.cmu-eats: Refreshing state... [id=16807600] github_repository.devops_link_shortener: Refreshing state... [id=link-shortener] github_team.tartan-vote: Refreshing state... [id=17701833] data.external.identity_brightz: Reading... data.external.identity_brightz: Read complete after 0s [id=-] data.external.identity_thesuperrl: Reading... data.external.identity_riztofu: Reading... data.external.identity_thesuperrl: Read complete after 0s [id=-] data.external.identity_xboxbedrock: Reading... data.external.identity_riztofu: Read complete after 0s [id=-] data.external.identity_gostmeaper: Reading... data.external.identity_jc_965: Reading... data.external.identity_xboxbedrock: Read complete after 1s [id=-] data.external.identity_anish: Reading... data.external.identity_gostmeaper: Read complete after 1s [id=-] data.external.identity_kritdass: Reading... data.external.identity_amzoeee: Reading... data.external.identity_jc_965: Read complete after 0s [id=-] data.external.identity_jiyas: Reading... data.external.identity_anish: Read complete after 0s [id=-] data.external.identity_emmawong: Reading... data.external.identity_kritdass: Read complete after 0s [id=-] data.external.identity_dkou: Reading... data.external.identity_amzoeee: Read complete after 0s [id=-] data.external.identity_siddharthr: Reading... data.external.identity_jiyas: Read complete after 0s [id=-] data.external.identity_rhsen: Reading... data.external.identity_emmawong: Read complete after 0s [id=-] data.external.identity_krishsax: Reading... data.external.identity_dkou: Read complete after 0s [id=-] data.external.identity_zoexue: Reading... data.external.identity_siddharthr: Read complete after 0s [id=-] data.external.identity_kechezon: Reading... data.external.identity_rhsen: Read complete after 0s [id=-] data.external.identity_kothurvidur: Reading... data.external.identity_krishsax: Read complete after 0s [id=-] data.external.identity_manrajs: Reading... data.external.identity_zoexue: Read complete after 0s [id=-] data.external.identity_etashjha: Reading... data.external.identity_liamwalk: Reading... data.external.identity_kothurvidur: Read complete after 0s [id=-] data.external.identity_tentype: Reading... data.external.identity_kechezon: Read complete after 1s [id=-] data.external.identity_nikhilr: Reading... data.external.identity_manrajs: Read complete after 1s [id=-] data.external.identity_maybe_yiyi: Reading... data.external.identity_etashjha: Read complete after 1s [id=-] data.external.identity_jessica_lr_song: Reading... data.external.identity_liamwalk: Read complete after 1s [id=-] data.external.identity_jw8: Reading... data.external.identity_tentype: Read complete after 1s [id=-] data.external.identity_acchs: Reading... data.external.identity_nikhilr: Read complete after 0s [id=-] data.external.identity_jalenluorion: Reading... data.external.identity_maybe_yiyi: Read complete after 0s [id=-] data.external.identity_rzhu3: Reading... data.external.identity_jessica_lr_song: Read complete after 0s [id=-] data.external.identity_dcrodrig: Reading... random_password.kennel_webhook_secret: Refreshing state... [id=none] forgejo_team_member.cmu-maps_xboxbedrock: Refreshing state... forgejo_repository.cmu-courses_internet_archive: Refreshing state... [name=internet-archive] data.external.identity_jw8: Read complete after 0s [id=-] forgejo_repository.devops_collie: Refreshing state... [name=collie] data.forgejo_organization.this: Reading... data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs] forgejo_team_member.cmu-maps_krishsax: Refreshing state... forgejo_repository.tartan-vote_tartan_vote: Refreshing state... [name=tartan-vote] forgejo_repository.cmu-maps_maps: Refreshing state... [name=maps] data.external.identity_acchs: Read complete after 0s [id=-] forgejo_repository.cmu-insta_insta: Refreshing state... [name=insta] forgejo_repository.cmu-eats_dining_api: Refreshing state... [name=dining-api] forgejo_repository.slai_cmugpt_agent: Refreshing state... [name=cmugpt-agent] forgejo_repository.quest_quest: Refreshing state... [name=quest] forgejo_team_member.maps_leads_xboxbedrock: Refreshing state... forgejo_team_member.maps_leads_krishsax: Refreshing state... data.external.identity_rzhu3: Read complete after 0s [id=-] data.external.identity_jalenluorion: Read complete after 0s [id=-] forgejo_repository.slai_cmugpt_sms_surface: Refreshing state... [name=cmugpt-sms-surface] forgejo_repository.devops_link_shortener: Refreshing state... [name=link-shortener] forgejo_repository.uia_components: Refreshing state... [name=components] forgejo_repository.cbp_discord_verify: Refreshing state... [name=discord-verify] data.external.identity_dcrodrig: Read complete after 0s [id=-] forgejo_repository.cbp_bus_sign: Refreshing state... [name=bus-sign] forgejo_repository.uia_scottylabs_org: Refreshing state... [name=scottylabs-org] forgejo_repository.cmu-cal_cal: Refreshing state... [name=cal] forgejo_repository.lost-and-found_lost_and_found_v2: Refreshing state... [name=lost-and-found-v2] forgejo_repository.cbp_groupme_mirror: Refreshing state... [name=groupme-mirror] forgejo_repository.devops_kennel: Refreshing state... [name=kennel] forgejo_repository.terrier_terrier: Refreshing state... [name=terrier] forgejo_repository.cmu-housing_housing: Refreshing state... [name=housing] forgejo_repository.devops_infrastructure: Refreshing state... [name=infrastructure] forgejo_repository.cmu-eats_eats: Refreshing state... [name=eats] forgejo_repository.cbp_dalmatian: Refreshing state... [name=dalmatian] forgejo_repository.slai_cmugpt_surface: Refreshing state... [name=cmugpt-surface] forgejo_repository.devops_governance: Refreshing state... [name=governance] forgejo_repository.devops_documentation: Refreshing state... [name=documentation] forgejo_repository.cmu-courses_courses: Refreshing state... [name=courses] forgejo_repository.occ_occ: Refreshing state... [name=occ] github_actions_repository_permissions.devops_collie_actions: Refreshing state... [id=collie] github_actions_repository_permissions.devops_documentation_actions: Refreshing state... [id=documentation] forgejo_repository.slai_mcp_server: Refreshing state... [name=mcp-server] github_actions_repository_permissions.lost-and-found_lost_and_found_v2_actions: Refreshing state... [id=lost-and-found-v2] github_actions_repository_permissions.cmu-cal_cal_actions: Refreshing state... [id=cal] github_actions_repository_permissions.cbp_discord_verify_actions: Refreshing state... [id=discord-verify] github_actions_repository_permissions.cmu-insta_insta_actions: Refreshing state... [id=insta] github_actions_repository_permissions.cmu-courses_internet_archive_actions: Refreshing state... [id=internet-archive] github_actions_repository_permissions.slai_mcp_server_actions: Refreshing state... [id=mcp-server] github_actions_repository_permissions.slai_cmugpt_agent_actions: Refreshing state... [id=cmugpt-agent] github_actions_repository_permissions.devops_infrastructure_actions: Refreshing state... [id=infrastructure] github_actions_repository_permissions.cmu-housing_housing_actions: Refreshing state... [id=housing] github_actions_repository_permissions.cbp_groupme_mirror_actions: Refreshing state... [id=groupme-mirror] github_actions_repository_permissions.devops_governance_actions: Refreshing state... [id=governance] github_actions_repository_permissions.cmu-maps_maps_actions: Refreshing state... [id=maps] github_actions_repository_permissions.cbp_bus_sign_actions: Refreshing state... [id=bus-sign] github_actions_repository_permissions.terrier_terrier_actions: Refreshing state... [id=terrier] github_actions_repository_permissions.cmu-eats_dining_api_actions: Refreshing state... [id=dining-api] github_actions_repository_permissions.uia_components_actions: Refreshing state... [id=components] github_actions_repository_permissions.tartan-vote_tartan_vote_actions: Refreshing state... [id=tartan-vote] github_actions_repository_permissions.uia_scottylabs_org_actions: Refreshing state... [id=scottylabs-org] github_actions_repository_permissions.slai_cmugpt_sms_surface_actions: Refreshing state... [id=cmugpt-sms-surface] github_actions_repository_permissions.slai_cmugpt_surface_actions: Refreshing state... [id=cmugpt-surface] github_actions_repository_permissions.occ_occ_actions: Refreshing state... [id=occ] github_team_membership.quest_brightz: Refreshing state... [id=17330197:bright-zheng] github_actions_repository_permissions.cmu-courses_courses_actions: Refreshing state... [id=courses] github_team_membership.devops_thesuperrl: Refreshing state... [id=14706218:thesuperrl] github_team_membership.cmu-housing_xboxbedrock: Refreshing state... [id=17937919:xboxbedrock] github_team_membership.quest_xboxbedrock: Refreshing state... [id=17330197:xboxbedrock] github_team_membership.devops_xboxbedrock: Refreshing state... [id=14706218:xboxbedrock] github_team_membership.cmu-housing_gostmeaper: Refreshing state... [id=17937919:gostmeaper] github_actions_repository_permissions.cbp_dalmatian_actions: Refreshing state... [id=dalmatian] github_team_membership.slai_jc_965: Refreshing state... [id=17571328:jc-965] github_team_membership.devops_anish: Refreshing state... [id=14706218:ap-1] github_team_membership.cmu-courses_anish: Refreshing state... [id=3643907:ap-1] github_team_membership.occ_anish: Refreshing state... [id=19240310:ap-1] github_team_membership.cmu-maps_anish: Refreshing state... [id=10872360:ap-1] github_team_membership.quest_anish: Refreshing state... [id=17330197:ap-1] github_team_membership.terrier_anish: Refreshing state... [id=12028577:ap-1] github_team_membership.terrier_kritdass: Refreshing state... [id=12028577:kritdass] github_team_membership.cmu-housing_amzoeee: Refreshing state... [id=17937919:amzoeee] github_team_membership.slai_jiyas: Refreshing state... [id=17571328:jiya2301] github_team_membership.cmu-housing_emmawong: Refreshing state... [id=17937919:emw868] github_team_membership.terrier_dkou: Refreshing state... [id=12028577:sharkree] github_team_membership.cmu-housing_rhsen: Refreshing state... [id=17937919:rohansen08] github_team_membership.slai_krishsax: Refreshing state... [id=17571328:krishsaxena] github_team_membership.cmu-housing_zoexue: Refreshing state... [id=17937919:o-zox-o] github_team_membership.cmu-eats_anish: Refreshing state... [id=16807600:ap-1] github_team_membership.terrier_kothurvidur: Refreshing state... [id=12028577:kothurvidur] github_team_membership.quest_kechezon: Refreshing state... [id=17330197:kechezon] github_team_membership.cmu-housing_manrajs: Refreshing state... [id=17937919:manraj10] github_team_membership.terrier_etashjha: Refreshing state... [id=12028577:etashj] github_team_membership.cmu-courses_etashjha: Refreshing state... [id=3643907:etashj] github_team_membership.cmu-insta_etashjha: Refreshing state... [id=19240311:etashj] github_team_membership.cbp_liamwalk: Refreshing state... [id=17170099:pandaaured] github_team_membership.cmu-housing_tentype: Refreshing state... [id=17937919:tentype] github_team_membership.cbp_tentype: Refreshing state... [id=17170099:tentype] github_team_membership.cmu-housing_nikhilr: Refreshing state... [id=17937919:ecstaticpilot] github_team_membership.cbp_maybe_yiyi: Refreshing state... [id=17170099:maybe-yiyi] github_team_membership.cmu-housing_jessica_lr_song: Refreshing state... [id=17937919:jessica-lr-song] github_team_membership.tartan-vote_riztofu: Refreshing state... [id=17701833:riztofu] github_team_membership.tartan-vote_maybe_yiyi: Refreshing state... [id=17701833:maybe-yiyi] github_team_membership.quest_jw8: Refreshing state... [id=17330197:jeffreyyijiwang] restapi_object.cmu-courses_internet_archive_push_mirror: Refreshing state... [id=remote_mirror_58X__Yfs0bR] forgejo_team.groupme-mirror_leads: Refreshing state... [name=groupme-mirror-leads] forgejo_team.cmugpt-sms-surface_leads: Refreshing state... [name=cmugpt-sms-surface-leads] forgejo_team.cmugpt-agent_leads: Refreshing state... [name=cmugpt-agent-leads] forgejo_team.insta_leads: Refreshing state... [name=insta-leads] forgejo_team.lost-and-found: Refreshing state... [name=lost-and-found] forgejo_team.quest: Refreshing state... [name=quest] forgejo_team.tartan-vote_leads: Refreshing state... [name=tartan-vote-leads] forgejo_team.quest_leads: Refreshing state... [name=quest-leads] forgejo_team.devops_leads: Refreshing state... [name=devops-leads] forgejo_team.cmugpt-surface_leads: Refreshing state... [name=cmugpt-surface-leads] forgejo_team.eats_leads: Refreshing state... [name=eats-leads] forgejo_team.link-shortener_leads: Refreshing state... [name=link-shortener-leads] forgejo_team.documentation_leads: Refreshing state... [name=documentation-leads] forgejo_team.cmugpt-mcp-server_leads: Refreshing state... [name=cmugpt-mcp-server-leads] forgejo_team.cmu-courses: Refreshing state... [name=cmu-courses] forgejo_team.cmu-insta: Refreshing state... [name=cmu-insta] forgejo_team.cmu-housing: Refreshing state... [name=cmu-housing] forgejo_team.cal_leads: Refreshing state... [name=cal-leads] forgejo_team.terrier_leads: Refreshing state... [name=terrier-leads] forgejo_team.cmu-maps: Refreshing state... [name=cmu-maps] forgejo_team.housing_leads: Refreshing state... [name=housing-leads] forgejo_team.maps_leads: Refreshing state... [name=maps-leads] forgejo_team.bus-sign_leads: Refreshing state... [name=bus-sign-leads] forgejo_team.cmu-courses_leads: Refreshing state... [name=cmu-courses-leads] forgejo_team.uia: Refreshing state... [name=uia] forgejo_team.terrier: Refreshing state... [name=terrier] forgejo_team.occ: Refreshing state... [name=occ] forgejo_team.devops: Refreshing state... [name=devops] forgejo_team.cbp: Refreshing state... [name=cbp] forgejo_team.cmu-cal: Refreshing state... [name=cmu-cal] forgejo_team.slai: Refreshing state... [name=slai] forgejo_team.occ-main_leads: Refreshing state... [name=occ-main-leads] forgejo_team.courses_leads: Refreshing state... [name=courses-leads] forgejo_team.tartan-vote: Refreshing state... [name=tartan-vote] forgejo_team.discord-verify_leads: Refreshing state... [name=discord-verify-leads] forgejo_team.cmu-eats: Refreshing state... [name=cmu-eats] forgejo_team.dalmatian_leads: Refreshing state... [name=dalmatian-leads] restapi_object.devops_collie_push_mirror: Refreshing state... [id=remote_mirror_ljvVejNFXaS] github_team_membership.occ_acchs: Refreshing state... [id=19240310:acchs-at-cmu] restapi_object.tartan-vote_tartan_vote_push_mirror: Refreshing state... [id=remote_mirror_EUFq4sFSnH5] restapi_object.tartan-vote_tartan_vote_docs_webhook: Refreshing state... [id=25] restapi_object.cmu-maps_maps_push_mirror: Refreshing state... [id=remote_mirror_e4_FUnvoslJ] restapi_object.cmu-eats_dining_api_push_mirror: Refreshing state... [id=remote_mirror_a4MsQ5rO1_Z] restapi_object.cmu-insta_insta_push_mirror: Refreshing state... [id=remote_mirror_891nu6XQ1us] github_actions_repository_permissions.cmu-eats_eats_actions: Refreshing state... [id=eats] github_team_membership.slai_rzhu3: Refreshing state... [id=17571328:richard-31415] github_team_membership.cmu-cal_jalenluorion: Refreshing state... [id=15669112:jalenluorion] restapi_object.slai_cmugpt_agent_push_mirror: Refreshing state... [id=remote_mirror_kNY6rVcxdyr] github_team_membership.occ_dcrodrig: Refreshing state... [id=19240310:newhacker1746] restapi_object.slai_cmugpt_sms_surface_push_mirror: Refreshing state... [id=remote_mirror__fkBkASN2jA] restapi_object.quest_quest_push_mirror: Refreshing state... [id=remote_mirror_U9d9BPAon29] restapi_object.cbp_discord_verify_push_mirror: Refreshing state... [id=remote_mirror_3MV7Up6cY1X] restapi_object.cbp_bus_sign_push_mirror: Refreshing state... [id=remote_mirror_MkKxQ2vA9pc] restapi_object.devops_link_shortener_push_mirror: Refreshing state... [id=remote_mirror_cp8daMpt82v] restapi_object.uia_components_push_mirror: Refreshing state... [id=remote_mirror_Kwwa8TDnTgc] restapi_object.uia_scottylabs_org_push_mirror: Refreshing state... [id=remote_mirror_h6yN6QdS1lD] github_actions_repository_permissions.devops_kennel_actions: Refreshing state... [id=kennel] restapi_object.lost-and-found_lost_and_found_v2_push_mirror: Refreshing state... [id=remote_mirror_n_UZBFU-CY4] restapi_object.cmu-cal_cal_push_mirror: Refreshing state... [id=remote_mirror_CO243uIONoq] restapi_object.terrier_terrier_push_mirror: Refreshing state... [id=remote_mirror_YopDgZksZv5] restapi_object.cmu-housing_housing_push_mirror: Refreshing state... [id=remote_mirror_uoV2HG4aiOT] restapi_object.cmu-housing_housing_docs_webhook: Refreshing state... [id=15] restapi_object.devops_infrastructure_push_mirror: Refreshing state... [id=remote_mirror_oYtFpKVzTaZ] restapi_object.cbp_groupme_mirror_push_mirror: Refreshing state... [id=remote_mirror_U61HJHB9VGE] restapi_object.devops_kennel_push_mirror: Refreshing state... [id=remote_mirror_cT9psJkhTQ4] restapi_object.cmu-eats_eats_push_mirror: Refreshing state... [id=remote_mirror_sjYwMA6NPgh] restapi_object.cbp_dalmatian_push_mirror: Refreshing state... [id=remote_mirror_qzlCpLK153F] restapi_object.cbp_dalmatian_docs_webhook: Refreshing state... [id=7] restapi_object.slai_cmugpt_surface_push_mirror: Refreshing state... [id=remote_mirror_cPTQN6S5SpB] restapi_object.devops_documentation_push_mirror: Refreshing state... [id=remote_mirror_izmsqaNB6Wx] restapi_object.devops_governance_push_mirror: Refreshing state... [id=remote_mirror_lt9JzkX1jMq] restapi_object.cmu-courses_courses_push_mirror: Refreshing state... [id=remote_mirror_WMR9yQwKFDF] restapi_object.slai_mcp_server_push_mirror: Refreshing state... [id=remote_mirror_Tqh2y0R_vDp] restapi_object.occ_occ_push_mirror: Refreshing state... [id=remote_mirror_moBLq1YG1qz] github_actions_repository_permissions.quest_quest_actions: Refreshing state... [id=quest] github_actions_repository_permissions.devops_link_shortener_actions: Refreshing state... [id=link-shortener] restapi_object.uia_scottylabs_org_kennel_webhook: Refreshing state... [id=29] restapi_object.devops_kennel_kennel_webhook: Refreshing state... [id=14] restapi_object.devops_infrastructure_kennel_webhook: Refreshing state... [id=26] restapi_object.slai_cmugpt_surface_kennel_webhook: Refreshing state... [id=23] restapi_object.slai_cmugpt_sms_surface_kennel_webhook: Refreshing state... [id=19] restapi_object.tartan-vote_tartan_vote_kennel_webhook: Refreshing state... [id=24] restapi_object.cmu-housing_housing_kennel_webhook: Refreshing state... [id=16] restapi_object.cmu-courses_courses_kennel_webhook: Refreshing state... [id=11] restapi_object.devops_collie_kennel_webhook: Refreshing state... [id=17] restapi_object.cbp_dalmatian_kennel_webhook: Refreshing state... [id=6] restapi_object.slai_cmugpt_agent_kennel_webhook: Refreshing state... [id=28] restapi_object.cmu-eats_eats_kennel_webhook: Refreshing state... [id=31] restapi_object.quest_quest_kennel_webhook: Refreshing state... [id=10] restapi_object.devops_link_shortener_kennel_webhook: Refreshing state... [id=8] restapi_object.cbp_discord_verify_kennel_webhook: Refreshing state... [id=18] restapi_object.cmu-cal_cal_kennel_webhook: Refreshing state... [id=20] restapi_object.terrier_terrier_kennel_webhook: Refreshing state... [id=12] restapi_object.slai_mcp_server_kennel_webhook: Refreshing state... [id=13] restapi_object.devops_governance_kennel_webhook: Refreshing state... [id=22] restapi_object.cbp_bus_sign_kennel_webhook: Refreshing state... [id=21] restapi_object.uia_components_kennel_webhook: Refreshing state... [id=27] restapi_object.cmu-maps_maps_kennel_webhook: Refreshing state... [id=9] github_repository_deploy_key.cmu-courses_internet_archive_mirror_deploy_key: Refreshing state... [id=internet-archive:159665356] forgejo_team_member.groupme-mirror_leads_tentype: Refreshing state... forgejo_team_member.groupme-mirror_leads_maybe_yiyi: Refreshing state... forgejo_team_member.cmugpt-sms-surface_leads_krishsax: Refreshing state... forgejo_team_member.cmugpt-agent_leads_krishsax: Refreshing state... forgejo_team_member.insta_leads_etashjha: Refreshing state... forgejo_team_repository.lost-and-found_lost_and_found_v2: Refreshing state... forgejo_team_member.quest_brightz: Refreshing state... forgejo_team_member.quest_jw8: Refreshing state... forgejo_team_repository.quest_quest: Refreshing state... forgejo_team_member.quest_xboxbedrock: Refreshing state... forgejo_team_member.quest_anish: Refreshing state... forgejo_team_member.quest_kechezon: Refreshing state... forgejo_team_member.tartan-vote_leads_maybe_yiyi: Refreshing state... forgejo_team_member.quest_leads_xboxbedrock: Refreshing state... forgejo_team_member.quest_leads_anish: Refreshing state... forgejo_team_member.devops_leads_thesuperrl: Refreshing state... forgejo_team_member.devops_leads_anish: Refreshing state... forgejo_team_member.cmugpt-surface_leads_krishsax: Refreshing state... forgejo_team_member.eats_leads_anish: Refreshing state... forgejo_team_member.link-shortener_leads_thesuperrl: Refreshing state... forgejo_team_member.link-shortener_leads_anish: Refreshing state... forgejo_team_member.documentation_leads_thesuperrl: Refreshing state... forgejo_team_member.documentation_leads_anish: Refreshing state... forgejo_team_member.cmugpt-mcp-server_leads_krishsax: Refreshing state... forgejo_team_member.cmu-courses_anish: Refreshing state... forgejo_team_repository.cmu-courses_courses: Refreshing state... forgejo_team_member.cmu-courses_etashjha: Refreshing state... forgejo_team_repository.cmu-courses_internet_archive: Refreshing state... forgejo_team_repository.cmu-insta_insta: Refreshing state... forgejo_team_member.cmu-insta_etashjha: Refreshing state... forgejo_team_member.cmu-housing_emmawong: Refreshing state... forgejo_team_member.cmu-housing_xboxbedrock: Refreshing state... forgejo_team_member.cmu-housing_gostmeaper: Refreshing state... forgejo_team_member.cmu-housing_manrajs: Refreshing state... forgejo_team_member.cmu-housing_nikhilr: Refreshing state... forgejo_team_member.cmu-housing_jessica_lr_song: Refreshing state... forgejo_team_member.cmu-housing_rhsen: Refreshing state... forgejo_team_member.cmu-housing_zoexue: Refreshing state... forgejo_team_member.cmu-housing_amzoeee: Refreshing state... forgejo_team_repository.cmu-housing_housing: Refreshing state... forgejo_team_member.cmu-housing_tentype: Refreshing state... forgejo_team_member.cal_leads_jalenluorion: Refreshing state... forgejo_team_member.terrier_leads_kritdass: Refreshing state... forgejo_team_member.terrier_leads_anish: Refreshing state... forgejo_team_repository.cmu-maps_maps: Refreshing state... forgejo_team_member.cmu-maps_anish: Refreshing state... forgejo_team_member.housing_leads_emmawong: Refreshing state... forgejo_team_member.housing_leads_gostmeaper: Refreshing state... forgejo_team_member.housing_leads_nikhilr: Refreshing state... forgejo_team_member.maps_leads_anish: Refreshing state... forgejo_team_member.housing_leads_tentype: Refreshing state... forgejo_team_member.bus-sign_leads_tentype: Refreshing state... forgejo_team_member.bus-sign_leads_maybe_yiyi: Refreshing state... forgejo_team_member.cmu-courses_leads_etashjha: Refreshing state... forgejo_team_member.cmu-courses_leads_anish: Refreshing state... forgejo_team_repository.uia_components: Refreshing state... forgejo_team_repository.uia_scottylabs_org: Refreshing state... forgejo_team_member.devops_anish: Refreshing state... forgejo_team_member.devops_xboxbedrock: Refreshing state... forgejo_team_repository.devops_link_shortener: Refreshing state... forgejo_team_repository.devops_governance: Refreshing state... forgejo_team_repository.devops_collie: Refreshing state... forgejo_team_repository.devops_documentation: Refreshing state... forgejo_team_repository.devops_kennel: Refreshing state... forgejo_team_repository.devops_infrastructure: Refreshing state... forgejo_team_member.devops_thesuperrl: Refreshing state... forgejo_team_member.terrier_etashjha: Refreshing state... forgejo_team_member.terrier_anish: Refreshing state... forgejo_team_repository.terrier_terrier: Refreshing state... forgejo_team_member.terrier_kritdass: Refreshing state... forgejo_team_member.terrier_dkou: Refreshing state... forgejo_team_member.terrier_kothurvidur: Refreshing state... forgejo_team_member.occ_dcrodrig: Refreshing state... forgejo_team_member.occ_acchs: Refreshing state... forgejo_team_repository.occ_occ: Refreshing state... forgejo_team_member.occ_anish: Refreshing state... forgejo_team_repository.cbp_discord_verify: Refreshing state... forgejo_team_member.cbp_liamwalk: Refreshing state... forgejo_team_member.cbp_maybe_yiyi: Refreshing state... forgejo_team_repository.cbp_bus_sign: Refreshing state... forgejo_team_repository.cbp_groupme_mirror: Refreshing state... forgejo_team_member.cbp_tentype: Refreshing state... forgejo_team_repository.cbp_dalmatian: Refreshing state... forgejo_team_repository.cmu-cal_cal: Refreshing state... forgejo_team_member.cmu-cal_jalenluorion: Refreshing state... forgejo_team_repository.slai_cmugpt_sms_surface: Refreshing state... forgejo_team_repository.slai_cmugpt_surface: Refreshing state... forgejo_team_repository.slai_cmugpt_agent: Refreshing state... forgejo_team_repository.slai_mcp_server: Refreshing state... forgejo_team_member.slai_jc_965: Refreshing state... forgejo_team_member.slai_jiyas: Refreshing state... forgejo_team_member.slai_rzhu3: Refreshing state... forgejo_team_member.slai_krishsax: Refreshing state... forgejo_team_member.occ-main_leads_dcrodrig: Refreshing state... forgejo_team_member.occ-main_leads_anish: Refreshing state... forgejo_team_member.courses_leads_etashjha: Refreshing state... forgejo_team_member.courses_leads_anish: Refreshing state... forgejo_team_member.tartan-vote_maybe_yiyi: Refreshing state... forgejo_team_repository.tartan-vote_tartan_vote: Refreshing state... forgejo_team_member.tartan-vote_riztofu: Refreshing state... forgejo_team_member.discord-verify_leads_maybe_yiyi: Refreshing state... forgejo_team_member.discord-verify_leads_tentype: Refreshing state... forgejo_team_repository.cmu-eats_eats: Refreshing state... forgejo_team_repository.cmu-eats_dining_api: Refreshing state... forgejo_team_member.cmu-eats_anish: Refreshing state... forgejo_team_member.dalmatian_leads_maybe_yiyi: Refreshing state... forgejo_team_member.dalmatian_leads_tentype: Refreshing state... github_repository_deploy_key.devops_collie_mirror_deploy_key: Refreshing state... [id=collie:159665332] github_repository_deploy_key.tartan-vote_tartan_vote_mirror_deploy_key: Refreshing state... [id=tartan-vote:159665350] github_repository_deploy_key.cmu-maps_maps_mirror_deploy_key: Refreshing state... [id=maps:159665325] github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Refreshing state... [id=dining-api:161772815] github_repository_deploy_key.cmu-insta_insta_mirror_deploy_key: Refreshing state... [id=insta:161720245] github_repository_deploy_key.slai_cmugpt_sms_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-sms-surface:159665365] github_repository_deploy_key.slai_cmugpt_agent_mirror_deploy_key: Refreshing state... [id=cmugpt-agent:159665360] github_repository_deploy_key.quest_quest_mirror_deploy_key: Refreshing state... [id=quest:159665323] github_repository_deploy_key.cbp_discord_verify_mirror_deploy_key: Refreshing state... [id=discord-verify:159665340] github_repository_deploy_key.cbp_bus_sign_mirror_deploy_key: Refreshing state... [id=bus-sign:159665342] github_repository_deploy_key.devops_link_shortener_mirror_deploy_key: Refreshing state... [id=link-shortener:159665345] github_repository_deploy_key.uia_components_mirror_deploy_key: Refreshing state... [id=components:159665353] github_repository_deploy_key.uia_scottylabs_org_mirror_deploy_key: Refreshing state... [id=scottylabs-org:159665361] github_repository_deploy_key.lost-and-found_lost_and_found_v2_mirror_deploy_key: Refreshing state... [id=lost-and-found-v2:159665337] github_repository_deploy_key.cmu-cal_cal_mirror_deploy_key: Refreshing state... [id=cal:159665344] github_repository_deploy_key.terrier_terrier_mirror_deploy_key: Refreshing state... [id=terrier:159665339] github_repository_deploy_key.cmu-housing_housing_mirror_deploy_key: Refreshing state... [id=housing:159665369] github_repository_deploy_key.devops_infrastructure_mirror_deploy_key: Refreshing state... [id=infrastructure:159665343] github_repository_deploy_key.cbp_groupme_mirror_mirror_deploy_key: Refreshing state... [id=groupme-mirror:159665366] github_repository_deploy_key.devops_kennel_mirror_deploy_key: Refreshing state... [id=kennel:159665328] github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Refreshing state... [id=eats:161775995] github_repository_deploy_key.cbp_dalmatian_mirror_deploy_key: Refreshing state... [id=dalmatian:159665367] github_repository_deploy_key.slai_cmugpt_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-surface:159665348] github_repository_deploy_key.devops_documentation_mirror_deploy_key: Refreshing state... [id=documentation:159665362] github_repository_deploy_key.devops_governance_mirror_deploy_key: Refreshing state... [id=governance:159665346] github_repository_deploy_key.cmu-courses_courses_mirror_deploy_key: Refreshing state... [id=courses:159665327] github_repository_deploy_key.slai_mcp_server_mirror_deploy_key: Refreshing state... [id=mcp-server:159665358] github_repository_deploy_key.occ_occ_mirror_deploy_key: Refreshing state... [id=occ:161720242] forgejo_team_repository.quest_leads_quest: Refreshing state... forgejo_team_repository.courses_leads_courses: Refreshing state... forgejo_team_repository.cmu-courses_leads_internet_archive: Refreshing state... forgejo_team_repository.insta_leads_insta: Refreshing state... forgejo_team_repository.housing_leads_housing: Refreshing state... forgejo_team_repository.maps_leads_maps: Refreshing state... forgejo_team_repository.devops_leads_governance: Refreshing state... forgejo_team_repository.devops_leads_infrastructure: Refreshing state... forgejo_team_repository.devops_leads_collie: Refreshing state... forgejo_team_repository.documentation_leads_documentation: Refreshing state... forgejo_team_repository.devops_leads_kennel: Refreshing state... forgejo_team_repository.link-shortener_leads_link_shortener: Refreshing state... forgejo_team_repository.terrier_leads_terrier: Refreshing state... forgejo_team_repository.occ-main_leads_occ: Refreshing state... forgejo_team_repository.cal_leads_cal: Refreshing state... forgejo_team_repository.discord-verify_leads_discord_verify: Refreshing state... forgejo_team_repository.bus-sign_leads_bus_sign: Refreshing state... forgejo_team_repository.dalmatian_leads_dalmatian: Refreshing state... forgejo_team_repository.cmugpt-sms-surface_leads_cmugpt_sms_surface: Refreshing state... forgejo_team_repository.cmugpt-mcp-server_leads_mcp_server: Refreshing state... forgejo_team_repository.groupme-mirror_leads_groupme_mirror: Refreshing state... forgejo_team_repository.cmugpt-agent_leads_cmugpt_agent: Refreshing state... forgejo_team_repository.cmugpt-surface_leads_cmugpt_surface: Refreshing state... forgejo_team_repository.tartan-vote_leads_tartan_vote: Refreshing state... forgejo_team_repository.eats_leads_eats: Refreshing state... forgejo_team_repository.eats_leads_dining_api: Refreshing state... github_repository_ruleset.cmu-courses_internet_archive_ruleset: Refreshing state... [id=11988074] github_repository_ruleset.devops_collie_ruleset: Refreshing state... [id=18865204] github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Refreshing state... [id=16848966] github_repository_ruleset.cmu-maps_maps_ruleset: Refreshing state... [id=18864961] github_repository_ruleset.cmu-eats_dining_api_ruleset: Refreshing state... [id=21879626] github_repository_ruleset.cmu-insta_insta_ruleset: Refreshing state... [id=21841636] github_repository_ruleset.slai_cmugpt_agent_ruleset: Refreshing state... [id=16476661] github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Refreshing state... [id=17653622] github_repository_ruleset.quest_quest_ruleset: Refreshing state... [id=15750052] github_repository_ruleset.cbp_discord_verify_ruleset: Refreshing state... [id=11449154] github_repository_ruleset.cbp_bus_sign_ruleset: Refreshing state... [id=15236550] github_repository_ruleset.devops_link_shortener_ruleset: Refreshing state... [id=17911647] github_repository_ruleset.uia_components_ruleset: Refreshing state... [id=16366085] github_repository_ruleset.uia_scottylabs_org_ruleset: Refreshing state... [id=18112214] github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Refreshing state... [id=18920618] github_repository_ruleset.cmu-cal_cal_ruleset: Refreshing state... [id=18919281] github_repository_ruleset.terrier_terrier_ruleset: Refreshing state... [id=12671003] github_repository_ruleset.cmu-housing_housing_ruleset: Refreshing state... [id=17421604] github_repository_ruleset.devops_infrastructure_ruleset: Refreshing state... [id=11275118] github_repository_ruleset.cbp_groupme_mirror_ruleset: Refreshing state... [id=12503296] github_repository_ruleset.devops_kennel_ruleset: Refreshing state... [id=13193453] github_repository_ruleset.cmu-eats_eats_ruleset: Refreshing state... [id=21330218] github_repository_ruleset.cbp_dalmatian_ruleset: Refreshing state... [id=17238915] github_repository_ruleset.slai_cmugpt_surface_ruleset: Refreshing state... [id=16476662] github_repository_ruleset.devops_documentation_ruleset: Refreshing state... [id=17079368] github_repository_ruleset.devops_governance_ruleset: Refreshing state... [id=14488792] github_repository_ruleset.cmu-courses_courses_ruleset: Refreshing state... [id=15855161] github_repository_ruleset.slai_mcp_server_ruleset: Refreshing state... [id=17082154] github_repository_ruleset.occ_occ_ruleset: Refreshing state... [id=21841634] OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ~ update in-place (current -> planned) - destroy OpenTofu planned the following actions, but then encountered a problem: # forgejo_team_member.cmu-maps_siddharthr will be created + resource "forgejo_team_member" "cmu-maps_siddharthr" { + team_id = 13 + user = "sradhak2" } # forgejo_team_member.cmu-maps_xboxbedrock will be destroyed # (because forgejo_team_member.cmu-maps_xboxbedrock is not in configuration) - resource "forgejo_team_member" "cmu-maps_xboxbedrock" { - team_id = 13 -> null - user = "xboxbedrock" -> null } # forgejo_team_member.maps_leads_krishsax will be destroyed # (because forgejo_team_member.maps_leads_krishsax is not in configuration) - resource "forgejo_team_member" "maps_leads_krishsax" { - team_id = 38 -> null - user = "krishsax" -> null } # forgejo_team_member.maps_leads_siddharthr will be created + resource "forgejo_team_member" "maps_leads_siddharthr" { + team_id = 38 + user = "sradhak2" } # forgejo_team_member.maps_leads_xboxbedrock will be destroyed # (because forgejo_team_member.maps_leads_xboxbedrock is not in configuration) - resource "forgejo_team_member" "maps_leads_xboxbedrock" { - team_id = 38 -> null - user = "xboxbedrock" -> null } # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place ~ resource "github_repository_ruleset" "cmu-eats_eats_ruleset" { id = "21330218" name = "Default" # (6 unchanged attributes hidden) ~ bypass_actors { ~ actor_id = 0 -> 161775995 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_team_membership.cmu-maps_krishsax will be destroyed # (because github_team_membership.cmu-maps_krishsax is not in configuration) - resource "github_team_membership" "cmu-maps_krishsax" { - etag = "W/\"0ca81c85b299f01891dc704fc1e1f11d9b5e73a022c43d28f0593b116aa3098f\"" -> null - id = "10872360:krishsaxena" -> null - role = "maintainer" -> null - team_id = "10872360" -> null - username = "krishsaxena" -> null } # github_team_membership.cmu-maps_siddharthr will be created + resource "github_team_membership" "cmu-maps_siddharthr" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = "10872360" + username = "overthrowing" } # github_team_membership.cmu-maps_xboxbedrock will be destroyed # (because github_team_membership.cmu-maps_xboxbedrock is not in configuration) - resource "github_team_membership" "cmu-maps_xboxbedrock" { - etag = "W/\"695105377afc546ab133d841be4a870c042dca48a6172b3388875744e9e1aef9\"" -> null - id = "10872360:xboxbedrock" -> null - role = "maintainer" -> null - team_id = "10872360" -> null - username = "xboxbedrock" -> null } Plan: 3 to add, 1 to change, 5 to destroy. ╷ │ Error: Unable to read team member │ │ User 'krishsax' in team with ID 13 not found: The target couldn't be found. ╵ ``` </details> --- ### Plan Summary 6 projects, 5 with changes, 0 with no changes, 1 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```

Apply Error

plan for dir "tofu/git" workspace "default" project "git" has status "plan_errored" and cannot be applied; run `atlantis plan`
**Apply Error** ``` plan for dir "tofu/git" workspace "default" project "git" has status "plan_errored" and cannot be applied; run `atlantis plan` ```
Owner

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.28s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-maps_ch0_krishsax will be destroyed
  # (because null_resource.slack_cmu-maps_ch0_krishsax is not in configuration)
- resource "null_resource" "slack_cmu-maps_ch0_krishsax" {
-       id       = "8867766601843827684" -> null
-       triggers = {
          - "channel" = "C05T7QRN4LF"
          - "user"    = "U09DL6NH8FL"
        } -> null
    }

  # null_resource.slack_cmu-maps_ch0_siddharthr will be created
+ resource "null_resource" "slack_cmu-maps_ch0_siddharthr" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C05T7QRN4LF"
          + "user"    = "U09DC6MFB5E"
        }
    }

  # null_resource.slack_cmu-maps_ch0_xboxbedrock will be destroyed
  # (because null_resource.slack_cmu-maps_ch0_xboxbedrock is not in configuration)
- resource "null_resource" "slack_cmu-maps_ch0_xboxbedrock" {
-       id       = "2901886759953880290" -> null
-       triggers = {
          - "channel" = "C05T7QRN4LF"
          - "user"    = "U0AQRKYLGHK"
        } -> null
    }

  # null_resource.slack_hub_siddharthr will be created
+ resource "null_resource" "slack_hub_siddharthr" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U09DC6MFB5E"
        }
    }

  # null_resource.slack_leads_siddharthr will be created
+ resource "null_resource" "slack_leads_siddharthr" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0AN1M0HDV0"
          + "user"    = "U09DC6MFB5E"
        }
    }

Plan: 3 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 3 to add, 0 to change, 2 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.36s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # googleplay_app_iam.maps_krishsax will be destroyed
  # (because googleplay_app_iam.maps_krishsax is not in configuration)
- resource "googleplay_app_iam" "maps_krishsax" {
-       app_id               = "com.cmumaps.app" -> null
-       expanded_permissions = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       permissions          = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       user_id              = "krishsax@andrew.cmu.edu" -> null
    }

  # googleplay_app_iam.maps_siddharthr will be created
+ resource "googleplay_app_iam" "maps_siddharthr" {
+       app_id               = "com.cmumaps.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_APP_CONTENT",
+           "CAN_MANAGE_PUBLIC_APKS",
+           "CAN_MANAGE_PUBLIC_LISTING",
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_REPLY_TO_REVIEWS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "sradhak2@andrew.cmu.edu"
    }

  # googleplay_app_iam.maps_xboxbedrock will be destroyed
  # (because googleplay_app_iam.maps_xboxbedrock is not in configuration)
- resource "googleplay_app_iam" "maps_xboxbedrock" {
-       app_id               = "com.cmumaps.app" -> null
-       expanded_permissions = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       permissions          = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       user_id              = "spola2@andrew.cmu.edu" -> null
    }

  # googleplay_user.krishsax will be destroyed
  # (because googleplay_user.krishsax is not in configuration)
- resource "googleplay_user" "krishsax" {
-       email                = "krishsax@andrew.cmu.edu" -> null
-       expanded_permissions = [
-           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ] -> null
-       global_permissions   = [
-           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ] -> null
-       id                   = "krishsax@andrew.cmu.edu" -> null
-       name                 = "developers/7984360456919420393/users/krishsax@andrew.cmu.edu" -> null
    }

  # googleplay_user.siddharthr will be created
+ resource "googleplay_user" "siddharthr" {
+       email                = "sradhak2@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

Plan: 2 to add, 0 to change, 3 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 2 to add, 0 to change, 3 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.23s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.ops_siddharthr will be created
+ resource "google_cloud_identity_group_membership" "ops_siddharthr" {
+       create_time = (known after apply)
+       group       = "groups/03fwokq036zsp1f"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "sradhak2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

  # google_cloud_identity_group_membership.tech_siddharthr will be created
+ resource "google_cloud_identity_group_membership" "tech_siddharthr" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "sradhak2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 2 to add, 0 to change, 0 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.14s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-maps_maps will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-maps_maps" {
        id       = "scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64"
!       members  = [
-           "krishsax",
-           "spola2",
+           "sradhak2",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-maps_maps_admins will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-maps_maps_admins" {
        id       = "scottylabs/group-memberships/75db0c7f-7cfb-4d89-b84f-0f50523ed55b"
!       members  = [
-           "krishsax",
-           "spola2",
+           "sradhak2",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "sradhak2",
            # (29 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 3 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 3 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.27s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # discord_member_roles.krishsax will be updated in-place
! resource "discord_member_roles" "krishsax" {
        id        = "1353118209024331948:782966391321395240"
        # (2 unchanged attributes hidden)

-       role {
-           has_role = true -> null
-           role_id  = "1526003694708195448" -> null
        }

        # (3 unchanged blocks hidden)
    }

  # discord_member_roles.siddharthr will be created
+ resource "discord_member_roles" "siddharthr" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "584841698186756167"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1519792320755662928"
        }
+       role {
+           has_role = true
+           role_id  = "1526003694708195448"
        }
    }

  # discord_member_roles.xboxbedrock will be updated in-place
! resource "discord_member_roles" "xboxbedrock" {
        id        = "1353118209024331948:712398210345205871"
        # (2 unchanged attributes hidden)

-       role {
-           has_role = true -> null
-           role_id  = "1526003694708195448" -> null
        }

        # (5 unchanged blocks hidden)
    }

Plan: 1 to add, 2 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 2 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.29s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)
- destroy

OpenTofu will perform the following actions:

  # forgejo_team_member.cmu-maps_krishsax will be destroyed
  # (because forgejo_team_member.cmu-maps_krishsax is not in configuration)
- resource "forgejo_team_member" "cmu-maps_krishsax" {
-       team_id = 13 -> null
-       user    = "krishsax" -> null
    }

  # forgejo_team_member.cmu-maps_siddharthr will be created
+ resource "forgejo_team_member" "cmu-maps_siddharthr" {
+       team_id = 13
+       user    = "sradhak2"
    }

  # forgejo_team_member.cmu-maps_xboxbedrock will be destroyed
  # (because forgejo_team_member.cmu-maps_xboxbedrock is not in configuration)
- resource "forgejo_team_member" "cmu-maps_xboxbedrock" {
-       team_id = 13 -> null
-       user    = "xboxbedrock" -> null
    }

  # forgejo_team_member.maps_leads_krishsax will be destroyed
  # (because forgejo_team_member.maps_leads_krishsax is not in configuration)
- resource "forgejo_team_member" "maps_leads_krishsax" {
-       team_id = 38 -> null
-       user    = "krishsax" -> null
    }

  # forgejo_team_member.maps_leads_siddharthr will be created
+ resource "forgejo_team_member" "maps_leads_siddharthr" {
+       team_id = 38
+       user    = "sradhak2"
    }

  # forgejo_team_member.maps_leads_xboxbedrock will be destroyed
  # (because forgejo_team_member.maps_leads_xboxbedrock is not in configuration)
- resource "forgejo_team_member" "maps_leads_xboxbedrock" {
-       team_id = 38 -> null
-       user    = "xboxbedrock" -> null
    }

  # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place
! resource "github_repository_ruleset" "cmu-eats_eats_ruleset" {
        id          = "21330218"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 161775995
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_team_membership.cmu-maps_krishsax will be destroyed
  # (because github_team_membership.cmu-maps_krishsax is not in configuration)
- resource "github_team_membership" "cmu-maps_krishsax" {
-       etag     = "W/\"0ca81c85b299f01891dc704fc1e1f11d9b5e73a022c43d28f0593b116aa3098f\"" -> null
-       id       = "10872360:krishsaxena" -> null
-       role     = "maintainer" -> null
-       team_id  = "10872360" -> null
-       username = "krishsaxena" -> null
    }

  # github_team_membership.cmu-maps_siddharthr will be created
+ resource "github_team_membership" "cmu-maps_siddharthr" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "maintainer"
+       team_id  = "10872360"
+       username = "overthrowing"
    }

  # github_team_membership.cmu-maps_xboxbedrock will be destroyed
  # (because github_team_membership.cmu-maps_xboxbedrock is not in configuration)
- resource "github_team_membership" "cmu-maps_xboxbedrock" {
-       etag     = "W/\"695105377afc546ab133d841be4a870c042dca48a6172b3388875744e9e1aef9\"" -> null
-       id       = "10872360:xboxbedrock" -> null
-       role     = "maintainer" -> null
-       team_id  = "10872360" -> null
-       username = "xboxbedrock" -> null
    }

Plan: 3 to add, 1 to change, 6 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 3 to add, 1 to change, 6 to destroy.


Plan Summary

6 projects, 6 with changes, 0 with no changes, 0 failed

  • To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.28s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # null_resource.slack_cmu-maps_ch0_krishsax will be destroyed # (because null_resource.slack_cmu-maps_ch0_krishsax is not in configuration) - resource "null_resource" "slack_cmu-maps_ch0_krishsax" { - id = "8867766601843827684" -> null - triggers = { - "channel" = "C05T7QRN4LF" - "user" = "U09DL6NH8FL" } -> null } # null_resource.slack_cmu-maps_ch0_siddharthr will be created + resource "null_resource" "slack_cmu-maps_ch0_siddharthr" { + id = (known after apply) + triggers = { + "channel" = "C05T7QRN4LF" + "user" = "U09DC6MFB5E" } } # null_resource.slack_cmu-maps_ch0_xboxbedrock will be destroyed # (because null_resource.slack_cmu-maps_ch0_xboxbedrock is not in configuration) - resource "null_resource" "slack_cmu-maps_ch0_xboxbedrock" { - id = "2901886759953880290" -> null - triggers = { - "channel" = "C05T7QRN4LF" - "user" = "U0AQRKYLGHK" } -> null } # null_resource.slack_hub_siddharthr will be created + resource "null_resource" "slack_hub_siddharthr" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U09DC6MFB5E" } } # null_resource.slack_leads_siddharthr will be created + resource "null_resource" "slack_leads_siddharthr" { + id = (known after apply) + triggers = { + "channel" = "C0AN1M0HDV0" + "user" = "U09DC6MFB5E" } } Plan: 3 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 3 to add, 0 to change, 2 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.36s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # googleplay_app_iam.maps_krishsax will be destroyed # (because googleplay_app_iam.maps_krishsax is not in configuration) - resource "googleplay_app_iam" "maps_krishsax" { - app_id = "com.cmumaps.app" -> null - expanded_permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - user_id = "krishsax@andrew.cmu.edu" -> null } # googleplay_app_iam.maps_siddharthr will be created + resource "googleplay_app_iam" "maps_siddharthr" { + app_id = "com.cmumaps.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_APP_CONTENT", + "CAN_MANAGE_PUBLIC_APKS", + "CAN_MANAGE_PUBLIC_LISTING", + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_REPLY_TO_REVIEWS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "sradhak2@andrew.cmu.edu" } # googleplay_app_iam.maps_xboxbedrock will be destroyed # (because googleplay_app_iam.maps_xboxbedrock is not in configuration) - resource "googleplay_app_iam" "maps_xboxbedrock" { - app_id = "com.cmumaps.app" -> null - expanded_permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - user_id = "spola2@andrew.cmu.edu" -> null } # googleplay_user.krishsax will be destroyed # (because googleplay_user.krishsax is not in configuration) - resource "googleplay_user" "krishsax" { - email = "krishsax@andrew.cmu.edu" -> null - expanded_permissions = [ - "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] -> null - global_permissions = [ - "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] -> null - id = "krishsax@andrew.cmu.edu" -> null - name = "developers/7984360456919420393/users/krishsax@andrew.cmu.edu" -> null } # googleplay_user.siddharthr will be created + resource "googleplay_user" "siddharthr" { + email = "sradhak2@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } Plan: 2 to add, 0 to change, 3 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 2 to add, 0 to change, 3 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.23s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.ops_siddharthr will be created + resource "google_cloud_identity_group_membership" "ops_siddharthr" { + create_time = (known after apply) + group = "groups/03fwokq036zsp1f" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "sradhak2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } # google_cloud_identity_group_membership.tech_siddharthr will be created + resource "google_cloud_identity_group_membership" "tech_siddharthr" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "sradhak2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.14s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-maps_maps will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-maps_maps" { id = "scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64" ! members = [ - "krishsax", - "spola2", + "sradhak2", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-maps_maps_admins will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-maps_maps_admins" { id = "scottylabs/group-memberships/75db0c7f-7cfb-4d89-b84f-0f50523ed55b" ! members = [ - "krishsax", - "spola2", + "sradhak2", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "sradhak2", # (29 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 3 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 3 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.27s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # discord_member_roles.krishsax will be updated in-place ! resource "discord_member_roles" "krishsax" { id = "1353118209024331948:782966391321395240" # (2 unchanged attributes hidden) - role { - has_role = true -> null - role_id = "1526003694708195448" -> null } # (3 unchanged blocks hidden) } # discord_member_roles.siddharthr will be created + resource "discord_member_roles" "siddharthr" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "584841698186756167" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1519792320755662928" } + role { + has_role = true + role_id = "1526003694708195448" } } # discord_member_roles.xboxbedrock will be updated in-place ! resource "discord_member_roles" "xboxbedrock" { id = "1353118209024331948:712398210345205871" # (2 unchanged attributes hidden) - role { - has_role = true -> null - role_id = "1526003694708195448" -> null } # (5 unchanged blocks hidden) } Plan: 1 to add, 2 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 2 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.29s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) - destroy OpenTofu will perform the following actions: # forgejo_team_member.cmu-maps_krishsax will be destroyed # (because forgejo_team_member.cmu-maps_krishsax is not in configuration) - resource "forgejo_team_member" "cmu-maps_krishsax" { - team_id = 13 -> null - user = "krishsax" -> null } # forgejo_team_member.cmu-maps_siddharthr will be created + resource "forgejo_team_member" "cmu-maps_siddharthr" { + team_id = 13 + user = "sradhak2" } # forgejo_team_member.cmu-maps_xboxbedrock will be destroyed # (because forgejo_team_member.cmu-maps_xboxbedrock is not in configuration) - resource "forgejo_team_member" "cmu-maps_xboxbedrock" { - team_id = 13 -> null - user = "xboxbedrock" -> null } # forgejo_team_member.maps_leads_krishsax will be destroyed # (because forgejo_team_member.maps_leads_krishsax is not in configuration) - resource "forgejo_team_member" "maps_leads_krishsax" { - team_id = 38 -> null - user = "krishsax" -> null } # forgejo_team_member.maps_leads_siddharthr will be created + resource "forgejo_team_member" "maps_leads_siddharthr" { + team_id = 38 + user = "sradhak2" } # forgejo_team_member.maps_leads_xboxbedrock will be destroyed # (because forgejo_team_member.maps_leads_xboxbedrock is not in configuration) - resource "forgejo_team_member" "maps_leads_xboxbedrock" { - team_id = 38 -> null - user = "xboxbedrock" -> null } # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place ! resource "github_repository_ruleset" "cmu-eats_eats_ruleset" { id = "21330218" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 161775995 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_team_membership.cmu-maps_krishsax will be destroyed # (because github_team_membership.cmu-maps_krishsax is not in configuration) - resource "github_team_membership" "cmu-maps_krishsax" { - etag = "W/\"0ca81c85b299f01891dc704fc1e1f11d9b5e73a022c43d28f0593b116aa3098f\"" -> null - id = "10872360:krishsaxena" -> null - role = "maintainer" -> null - team_id = "10872360" -> null - username = "krishsaxena" -> null } # github_team_membership.cmu-maps_siddharthr will be created + resource "github_team_membership" "cmu-maps_siddharthr" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = "10872360" + username = "overthrowing" } # github_team_membership.cmu-maps_xboxbedrock will be destroyed # (because github_team_membership.cmu-maps_xboxbedrock is not in configuration) - resource "github_team_membership" "cmu-maps_xboxbedrock" { - etag = "W/\"695105377afc546ab133d841be4a870c042dca48a6172b3388875744e9e1aef9\"" -> null - id = "10872360:xboxbedrock" -> null - role = "maintainer" -> null - team_id = "10872360" -> null - username = "xboxbedrock" -> null } Plan: 3 to add, 1 to change, 6 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 3 to add, 1 to change, 6 to destroy. --- ### Plan Summary 6 projects, 6 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis apply

atlantis apply

Ran Apply for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.43s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_cmu-maps_ch0_xboxbedrock: Destroying... [id=2901886759953880290]
null_resource.slack_cmu-maps_ch0_siddharthr: Creating...
null_resource.slack_leads_siddharthr: Creating...
null_resource.slack_leads_siddharthr: Provisioning with 'local-exec'...
null_resource.slack_leads_siddharthr (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0AN1M0HDV0 --user U09DC6MFB5E"]
null_resource.slack_cmu-maps_ch0_krishsax: Destroying... [id=8867766601843827684]
null_resource.slack_hub_siddharthr: Creating...
null_resource.slack_cmu-maps_ch0_xboxbedrock: Destruction complete after 0s
null_resource.slack_cmu-maps_ch0_siddharthr: Provisioning with 'local-exec'...
null_resource.slack_cmu-maps_ch0_siddharthr (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C05T7QRN4LF --user U09DC6MFB5E"]
null_resource.slack_cmu-maps_ch0_krishsax: Destruction complete after 0s
null_resource.slack_hub_siddharthr: Provisioning with 'local-exec'...
null_resource.slack_hub_siddharthr (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U09DC6MFB5E"]
null_resource.slack_hub_siddharthr: Creation complete after 1s [id=5103539595313716898]
null_resource.slack_cmu-maps_ch0_siddharthr: Creation complete after 1s [id=2628038141220469703]
null_resource.slack_leads_siddharthr: Creation complete after 1s [id=4246460775181380697]

Apply complete! Resources: 3 added, 0 changed, 2 destroyed.

2. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.34s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
googleplay_app_iam.maps_xboxbedrock: Destroying...
googleplay_app_iam.maps_krishsax: Destroying...
googleplay_user.siddharthr: Creating...
googleplay_user.siddharthr: Creation complete after 1s [id=sradhak2@andrew.cmu.edu]
googleplay_app_iam.maps_siddharthr: Creating...
googleplay_app_iam.maps_krishsax: Destruction complete after 1s
googleplay_user.krishsax: Destroying... [id=krishsax@andrew.cmu.edu]
googleplay_app_iam.maps_xboxbedrock: Destruction complete after 1s
googleplay_user.krishsax: Destruction complete after 0s
googleplay_app_iam.maps_siddharthr: Creation complete after 1s

Apply complete! Resources: 2 added, 0 changed, 3 destroyed.

3. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.47s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.siddharthr: Creating...
discord_member_roles.krishsax: Modifying... [id=1353118209024331948:782966391321395240]
discord_member_roles.xboxbedrock: Modifying... [id=1353118209024331948:712398210345205871]
discord_member_roles.krishsax: Modifications complete after 2s [id=1353118209024331948:782966391321395240]
discord_member_roles.xboxbedrock: Modifications complete after 2s [id=1353118209024331948:712398210345205871]
discord_member_roles.siddharthr: Creation complete after 3s [id=1353118209024331948:584841698186756167]

Apply complete! Resources: 1 added, 2 changed, 0 destroyed.

4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.32s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_siddharthr: Creating...
google_cloud_identity_group_membership.ops_siddharthr: Creating...
google_cloud_identity_group_membership.ops_siddharthr: Creation complete after 2s [id=groups/03fwokq036zsp1f/memberships/117208415664072816789]
google_cloud_identity_group_membership.tech_siddharthr: Creation complete after 2s [id=groups/04anzqyu1bqbv79/memberships/117208415664072816789]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

5. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.49s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cmu-maps_maps: Modifying... [id=scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64]
keycloak_group_memberships.project_cmu-maps_maps_admins: Modifying... [id=scottylabs/group-memberships/75db0c7f-7cfb-4d89-b84f-0f50523ed55b]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_cmu-maps_maps_admins: Modifications complete after 4s [id=scottylabs/group-memberships/75db0c7f-7cfb-4d89-b84f-0f50523ed55b]
keycloak_group_memberships.project_cmu-maps_maps: Modifications complete after 4s [id=scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64]
keycloak_group_memberships.projects: Modifications complete after 4s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 3 changed, 0 destroyed.

6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.36s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.maps_leads_xboxbedrock: Destroying...
forgejo_team_member.cmu-maps_krishsax: Destroying...
forgejo_team_member.maps_leads_krishsax: Destroying...
forgejo_team_member.cmu-maps_xboxbedrock: Destroying...
forgejo_team_member.maps_leads_siddharthr: Creating...
forgejo_team_member.cmu-maps_siddharthr: Creating...
forgejo_team_member.maps_leads_xboxbedrock: Destruction complete after 0s
forgejo_team_member.cmu-maps_krishsax: Destruction complete after 0s
forgejo_team_member.cmu-maps_xboxbedrock: Destruction complete after 0s
forgejo_team_member.maps_leads_krishsax: Destruction complete after 0s
forgejo_team_member.cmu-maps_siddharthr: Creation complete after 0s
forgejo_team_member.maps_leads_siddharthr: Creation complete after 0s
github_team_membership.cmu-maps_krishsax: Destroying... [id=10872360:krishsaxena]
github_team_membership.cmu-maps_xboxbedrock: Destroying... [id=10872360:xboxbedrock]
github_team_membership.cmu-maps_siddharthr: Creating...
github_repository_ruleset.cmu-eats_eats_ruleset: Modifying... [id=21330218]
github_team_membership.cmu-maps_xboxbedrock: Destruction complete after 0s
github_team_membership.cmu-maps_krishsax: Destruction complete after 1s
github_team_membership.cmu-maps_siddharthr: Creation complete after 4s [id=10872360:overthrowing]
github_repository_ruleset.cmu-eats_eats_ruleset: Modifications complete after 4s [id=21330218]

Apply complete! Resources: 3 added, 1 changed, 6 destroyed.

Apply Summary

6 projects, 6 successful, 0 failed, 0 errored

Ran Apply for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.43s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_cmu-maps_ch0_xboxbedrock: Destroying... [id=2901886759953880290] null_resource.slack_cmu-maps_ch0_siddharthr: Creating... null_resource.slack_leads_siddharthr: Creating... null_resource.slack_leads_siddharthr: Provisioning with 'local-exec'... null_resource.slack_leads_siddharthr (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0AN1M0HDV0 --user U09DC6MFB5E"] null_resource.slack_cmu-maps_ch0_krishsax: Destroying... [id=8867766601843827684] null_resource.slack_hub_siddharthr: Creating... null_resource.slack_cmu-maps_ch0_xboxbedrock: Destruction complete after 0s null_resource.slack_cmu-maps_ch0_siddharthr: Provisioning with 'local-exec'... null_resource.slack_cmu-maps_ch0_siddharthr (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C05T7QRN4LF --user U09DC6MFB5E"] null_resource.slack_cmu-maps_ch0_krishsax: Destruction complete after 0s null_resource.slack_hub_siddharthr: Provisioning with 'local-exec'... null_resource.slack_hub_siddharthr (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U09DC6MFB5E"] null_resource.slack_hub_siddharthr: Creation complete after 1s [id=5103539595313716898] null_resource.slack_cmu-maps_ch0_siddharthr: Creation complete after 1s [id=2628038141220469703] null_resource.slack_leads_siddharthr: Creation complete after 1s [id=4246460775181380697] Apply complete! Resources: 3 added, 0 changed, 2 destroyed. ``` </details> --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.34s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ googleplay_app_iam.maps_xboxbedrock: Destroying... googleplay_app_iam.maps_krishsax: Destroying... googleplay_user.siddharthr: Creating... googleplay_user.siddharthr: Creation complete after 1s [id=sradhak2@andrew.cmu.edu] googleplay_app_iam.maps_siddharthr: Creating... googleplay_app_iam.maps_krishsax: Destruction complete after 1s googleplay_user.krishsax: Destroying... [id=krishsax@andrew.cmu.edu] googleplay_app_iam.maps_xboxbedrock: Destruction complete after 1s googleplay_user.krishsax: Destruction complete after 0s googleplay_app_iam.maps_siddharthr: Creation complete after 1s Apply complete! Resources: 2 added, 0 changed, 3 destroyed. ``` </details> --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.47s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.siddharthr: Creating... discord_member_roles.krishsax: Modifying... [id=1353118209024331948:782966391321395240] discord_member_roles.xboxbedrock: Modifying... [id=1353118209024331948:712398210345205871] discord_member_roles.krishsax: Modifications complete after 2s [id=1353118209024331948:782966391321395240] discord_member_roles.xboxbedrock: Modifications complete after 2s [id=1353118209024331948:712398210345205871] discord_member_roles.siddharthr: Creation complete after 3s [id=1353118209024331948:584841698186756167] Apply complete! Resources: 1 added, 2 changed, 0 destroyed. ``` </details> --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.32s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_siddharthr: Creating... google_cloud_identity_group_membership.ops_siddharthr: Creating... google_cloud_identity_group_membership.ops_siddharthr: Creation complete after 2s [id=groups/03fwokq036zsp1f/memberships/117208415664072816789] google_cloud_identity_group_membership.tech_siddharthr: Creation complete after 2s [id=groups/04anzqyu1bqbv79/memberships/117208415664072816789] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.49s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cmu-maps_maps: Modifying... [id=scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64] keycloak_group_memberships.project_cmu-maps_maps_admins: Modifying... [id=scottylabs/group-memberships/75db0c7f-7cfb-4d89-b84f-0f50523ed55b] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_cmu-maps_maps_admins: Modifications complete after 4s [id=scottylabs/group-memberships/75db0c7f-7cfb-4d89-b84f-0f50523ed55b] keycloak_group_memberships.project_cmu-maps_maps: Modifications complete after 4s [id=scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64] keycloak_group_memberships.projects: Modifications complete after 4s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 3 changed, 0 destroyed. ``` </details> --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.36s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.maps_leads_xboxbedrock: Destroying... forgejo_team_member.cmu-maps_krishsax: Destroying... forgejo_team_member.maps_leads_krishsax: Destroying... forgejo_team_member.cmu-maps_xboxbedrock: Destroying... forgejo_team_member.maps_leads_siddharthr: Creating... forgejo_team_member.cmu-maps_siddharthr: Creating... forgejo_team_member.maps_leads_xboxbedrock: Destruction complete after 0s forgejo_team_member.cmu-maps_krishsax: Destruction complete after 0s forgejo_team_member.cmu-maps_xboxbedrock: Destruction complete after 0s forgejo_team_member.maps_leads_krishsax: Destruction complete after 0s forgejo_team_member.cmu-maps_siddharthr: Creation complete after 0s forgejo_team_member.maps_leads_siddharthr: Creation complete after 0s github_team_membership.cmu-maps_krishsax: Destroying... [id=10872360:krishsaxena] github_team_membership.cmu-maps_xboxbedrock: Destroying... [id=10872360:xboxbedrock] github_team_membership.cmu-maps_siddharthr: Creating... github_repository_ruleset.cmu-eats_eats_ruleset: Modifying... [id=21330218] github_team_membership.cmu-maps_xboxbedrock: Destruction complete after 0s github_team_membership.cmu-maps_krishsax: Destruction complete after 1s github_team_membership.cmu-maps_siddharthr: Creation complete after 4s [id=10872360:overthrowing] github_repository_ruleset.cmu-eats_eats_ruleset: Modifications complete after 4s [id=21330218] Apply complete! Resources: 3 added, 1 changed, 6 destroyed. ``` </details> --- ### Apply Summary 6 projects, 6 successful, 0 failed, 0 errored

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
Sign in to join this conversation.
No description provided.