feat: add SH to slai #458

Merged
xboxbedrock merged 3 commits from SH-main into main 2026-09-28 16:25:38 +00:00
Member

Replaces #450

Replaces #450
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 6 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: discord Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan 5/5 projects planned successfully.
atlantis/apply: discord Apply failed.
atlantis/apply: git Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply 1/5 projects failed to apply.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
8adffd138b
fix: governance
All checks were successful
CI / check-1 (pull_request) Successful in 6m56s
CI / check (pull_request) Successful in 0s
atlantis/plan: slack Plan: 0 to add, 0 to change, 2 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 2 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 0 to add, 0 to change, 1 to destroy.
atlantis/plan: google_play Plan: 0 to add, 0 to change, 2 to destroy.
Validate / Governance (pull_request_target) Successful in 5m38s
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: git Plan: 0 to add, 0 to change, 1 to destroy.
atlantis/plan: discord No changes. Your infrastructure matches the configuration.
atlantis/plan 6/6 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
CI / build (pull_request) Has been skipped
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
kennel/build skipped: no services or static sites defined
71f2359d52
xboxbedrock changed title from feat: add SH to sla to feat: add SH to slai 2026-09-28 16:18:40 +00:00

Plan Error

cannot run "plan": pull request 458 is currently locked by "plan" for commit 71f2359.
Wait until the previous command is complete and try again
**Plan Error** ``` cannot run "plan": pull request 458 is currently locked by "plan" for commit 71f2359. Wait until the previous command is complete and try again ``` - Commit: [71f2359d528feb46faac72933dded08d8845752e](https://git.cmu.dev/ScottyLabs/governance/pulls/458/commits/71f2359d528feb46faac72933dded08d8845752e) - Blocking job: [View Atlantis execution](https://atlantis.scottylabs.org/jobs/d1a02b91-15fa-4571-97a8-7e072eb2b44a)

Ran Plan for 6 projects:

  1. project: discord dir: tofu/discord workspace: default
  2. project: git dir: tofu/git workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: slack dir: tofu/slack workspace: default
  6. project: google_play dir: tofu/google_play workspace: default

1. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #450. To continue, delete the lock from #450 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #450. To continue, delete the lock from #450 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #450. To continue, delete the lock from #450 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #450. To continue, delete the lock from #450 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #450. To continue, delete the lock from #450 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: google_play dir: tofu/google_play workspace: default

Show Output
Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/458/default/crates/governance-schema)
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/458/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/458/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/458/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 8.66s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
- destroy

OpenTofu will perform the following actions:

  # googleplay_app_iam.eats_stanleymw will be destroyed
  # (because googleplay_app_iam.eats_stanleymw is not in configuration)
- resource "googleplay_app_iam" "eats_stanleymw" {
-       app_id               = "com.cmueats.app" -> null
-       expanded_permissions = [
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       permissions          = [
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       user_id              = "smwang2@andrew.cmu.edu" -> null
    }

  # googleplay_user.stanleymw will be destroyed
  # (because googleplay_user.stanleymw is not in configuration)
- resource "googleplay_user" "stanleymw" {
-       email                = "smwang2@andrew.cmu.edu" -> null
-       expanded_permissions = [
-           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ] -> null
-       global_permissions   = [
-           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ] -> null
-       id                   = "smwang2@andrew.cmu.edu" -> null
-       name                 = "developers/7984360456919420393/users/smwang2@andrew.cmu.edu" -> null
    }

Plan: 0 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 0 to add, 0 to change, 2 to destroy.


Plan Summary

6 projects, 1 with changes, 0 with no changes, 5 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` --- ### 1. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #450. To continue, delete the lock from #450 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #450. To continue, delete the lock from #450 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #450. To continue, delete the lock from #450 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #450. To continue, delete the lock from #450 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #450. To continue, delete the lock from #450 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/458/default/crates/governance-schema) Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/458/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/458/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/458/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 8.66s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: - destroy OpenTofu will perform the following actions: # googleplay_app_iam.eats_stanleymw will be destroyed # (because googleplay_app_iam.eats_stanleymw is not in configuration) - resource "googleplay_app_iam" "eats_stanleymw" { - app_id = "com.cmueats.app" -> null - expanded_permissions = [ - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - permissions = [ - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - user_id = "smwang2@andrew.cmu.edu" -> null } # googleplay_user.stanleymw will be destroyed # (because googleplay_user.stanleymw is not in configuration) - resource "googleplay_user" "stanleymw" { - email = "smwang2@andrew.cmu.edu" -> null - expanded_permissions = [ - "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] -> null - global_permissions = [ - "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] -> null - id = "smwang2@andrew.cmu.edu" -> null - name = "developers/7984360456919420393/users/smwang2@andrew.cmu.edu" -> null } Plan: 0 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 0 to add, 0 to change, 2 to destroy. --- ### Plan Summary 6 projects, 1 with changes, 0 with no changes, 5 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Author
Member

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: keycloak dir: tofu/keycloak workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: google_play dir: tofu/google_play workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: discord dir: tofu/discord workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.20s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-eats_ch0_stanleymw will be destroyed
  # (because null_resource.slack_cmu-eats_ch0_stanleymw is not in configuration)
- resource "null_resource" "slack_cmu-eats_ch0_stanleymw" {
-       id       = "5314915522024758797" -> null
-       triggers = {
          - "channel" = "C043UV30DHA"
          - "user"    = "U09DUH5550B"
        } -> null
    }

  # null_resource.slack_hub_stanleymw will be destroyed
  # (because null_resource.slack_hub_stanleymw is not in configuration)
- resource "null_resource" "slack_hub_stanleymw" {
-       id       = "7881991068013972679" -> null
-       triggers = {
          - "channel" = "C096TM8EMS8"
          - "user"    = "U09DUH5550B"
        } -> null
    }

Plan: 0 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 0 to add, 0 to change, 2 to destroy.


2. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.17s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-eats_eats will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-eats_eats" {
        id       = "scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba"
!       members  = [
-           "smwang2",
            # (7 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
-           "smwang2",
            # (103 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 2 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 2 to change, 0 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
- destroy

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_stanleymw will be destroyed
  # (because google_cloud_identity_group_membership.tech_stanleymw is not in configuration)
- resource "google_cloud_identity_group_membership" "tech_stanleymw" {
-       create_time = "2026-09-28T16:03:08.151110Z" -> null
-       group       = "groups/04anzqyu1bqbv79" -> null
-       id          = "groups/04anzqyu1bqbv79/memberships/108748397014985502908" -> null
-       name        = "groups/04anzqyu1bqbv79/memberships/108748397014985502908" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-28T16:03:08.151110Z" -> null

-       preferred_member_key {
-           id = "smwang2@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

Plan: 0 to add, 0 to change, 1 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 0 to add, 0 to change, 1 to destroy.


4. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
- destroy

OpenTofu will perform the following actions:

  # googleplay_app_iam.eats_stanleymw will be destroyed
  # (because googleplay_app_iam.eats_stanleymw is not in configuration)
- resource "googleplay_app_iam" "eats_stanleymw" {
-       app_id               = "com.cmueats.app" -> null
-       expanded_permissions = [
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       permissions          = [
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> null
-       user_id              = "smwang2@andrew.cmu.edu" -> null
    }

  # googleplay_user.stanleymw will be destroyed
  # (because googleplay_user.stanleymw is not in configuration)
- resource "googleplay_user" "stanleymw" {
-       email                = "smwang2@andrew.cmu.edu" -> null
-       expanded_permissions = [
-           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ] -> null
-       global_permissions   = [
-           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ] -> null
-       id                   = "smwang2@andrew.cmu.edu" -> null
-       name                 = "developers/7984360456919420393/users/smwang2@andrew.cmu.edu" -> null
    }

Plan: 0 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 0 to add, 0 to change, 2 to destroy.


5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.19s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
- destroy

OpenTofu will perform the following actions:

  # forgejo_team_member.cmu-eats_stanleymw will be destroyed
  # (because forgejo_team_member.cmu-eats_stanleymw is not in configuration)
- resource "forgejo_team_member" "cmu-eats_stanleymw" {
-       team_id = 28 -> null
-       user    = "stanleymw" -> null
    }

Plan: 0 to add, 0 to change, 1 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 0 to add, 0 to change, 1 to destroy.


6. project: discord dir: tofu/discord workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.16s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan Summary

6 projects, 5 with changes, 1 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.20s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: - destroy OpenTofu will perform the following actions: # null_resource.slack_cmu-eats_ch0_stanleymw will be destroyed # (because null_resource.slack_cmu-eats_ch0_stanleymw is not in configuration) - resource "null_resource" "slack_cmu-eats_ch0_stanleymw" { - id = "5314915522024758797" -> null - triggers = { - "channel" = "C043UV30DHA" - "user" = "U09DUH5550B" } -> null } # null_resource.slack_hub_stanleymw will be destroyed # (because null_resource.slack_hub_stanleymw is not in configuration) - resource "null_resource" "slack_hub_stanleymw" { - id = "7881991068013972679" -> null - triggers = { - "channel" = "C096TM8EMS8" - "user" = "U09DUH5550B" } -> null } Plan: 0 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 0 to add, 0 to change, 2 to destroy. --- ### 2. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.17s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-eats_eats will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-eats_eats" { id = "scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba" ! members = [ - "smwang2", # (7 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ - "smwang2", # (103 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 2 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 2 to change, 0 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: - destroy OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_stanleymw will be destroyed # (because google_cloud_identity_group_membership.tech_stanleymw is not in configuration) - resource "google_cloud_identity_group_membership" "tech_stanleymw" { - create_time = "2026-09-28T16:03:08.151110Z" -> null - group = "groups/04anzqyu1bqbv79" -> null - id = "groups/04anzqyu1bqbv79/memberships/108748397014985502908" -> null - name = "groups/04anzqyu1bqbv79/memberships/108748397014985502908" -> null - type = "USER" -> null - update_time = "2026-09-28T16:03:08.151110Z" -> null - preferred_member_key { - id = "smwang2@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } Plan: 0 to add, 0 to change, 1 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 0 to add, 0 to change, 1 to destroy. --- ### 4. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: - destroy OpenTofu will perform the following actions: # googleplay_app_iam.eats_stanleymw will be destroyed # (because googleplay_app_iam.eats_stanleymw is not in configuration) - resource "googleplay_app_iam" "eats_stanleymw" { - app_id = "com.cmueats.app" -> null - expanded_permissions = [ - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - permissions = [ - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> null - user_id = "smwang2@andrew.cmu.edu" -> null } # googleplay_user.stanleymw will be destroyed # (because googleplay_user.stanleymw is not in configuration) - resource "googleplay_user" "stanleymw" { - email = "smwang2@andrew.cmu.edu" -> null - expanded_permissions = [ - "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] -> null - global_permissions = [ - "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] -> null - id = "smwang2@andrew.cmu.edu" -> null - name = "developers/7984360456919420393/users/smwang2@andrew.cmu.edu" -> null } Plan: 0 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 0 to add, 0 to change, 2 to destroy. --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.19s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: - destroy OpenTofu will perform the following actions: # forgejo_team_member.cmu-eats_stanleymw will be destroyed # (because forgejo_team_member.cmu-eats_stanleymw is not in configuration) - resource "forgejo_team_member" "cmu-eats_stanleymw" { - team_id = 28 -> null - user = "stanleymw" -> null } Plan: 0 to add, 0 to change, 1 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 0 to add, 0 to change, 1 to destroy. --- ### 6. project: `discord` dir: `tofu/discord` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.16s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` --- ### Plan Summary 6 projects, 5 with changes, 1 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Author
Member

atlantis apply

atlantis apply

Ran Apply for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: git dir: tofu/git workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default
  6. project: google_play dir: tofu/google_play workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.20s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_cmu-eats_ch0_stanleymw: Destroying... [id=5314915522024758797]
null_resource.slack_hub_stanleymw: Destroying... [id=7881991068013972679]
null_resource.slack_hub_stanleymw: Destruction complete after 0s
null_resource.slack_cmu-eats_ch0_stanleymw: Destruction complete after 0s

Apply complete! Resources: 0 added, 0 changed, 2 destroyed.

2. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.22s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

3. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.cmu-eats_stanleymw: Destroying...
forgejo_team_member.cmu-eats_stanleymw: Destruction complete after 0s

Apply complete! Resources: 0 added, 0 changed, 1 destroyed.

4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_stanleymw: Destroying... [id=groups/04anzqyu1bqbv79/memberships/108748397014985502908]
google_cloud_identity_group_membership.tech_stanleymw: Destruction complete after 1s

Apply complete! Resources: 0 added, 0 changed, 1 destroyed.

5. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cmu-eats_eats: Modifying... [id=scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_cmu-eats_eats: Modifications complete after 0s [id=scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba]
keycloak_group_memberships.projects: Modifications complete after 2s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 2 changed, 0 destroyed.

6. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.23s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
googleplay_app_iam.eats_stanleymw: Destroying...
googleplay_app_iam.eats_stanleymw: Destruction complete after 3s
googleplay_user.stanleymw: Destroying... [id=smwang2@andrew.cmu.edu]
googleplay_user.stanleymw: Destruction complete after 0s

Apply complete! Resources: 0 added, 0 changed, 2 destroyed.

Apply Summary

6 projects, 6 successful, 0 failed, 0 errored

Ran Apply for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.20s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_cmu-eats_ch0_stanleymw: Destroying... [id=5314915522024758797] null_resource.slack_hub_stanleymw: Destroying... [id=7881991068013972679] null_resource.slack_hub_stanleymw: Destruction complete after 0s null_resource.slack_cmu-eats_ch0_stanleymw: Destruction complete after 0s Apply complete! Resources: 0 added, 0 changed, 2 destroyed. ``` </details> --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.22s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.cmu-eats_stanleymw: Destroying... forgejo_team_member.cmu-eats_stanleymw: Destruction complete after 0s Apply complete! Resources: 0 added, 0 changed, 1 destroyed. ``` </details> --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_stanleymw: Destroying... [id=groups/04anzqyu1bqbv79/memberships/108748397014985502908] google_cloud_identity_group_membership.tech_stanleymw: Destruction complete after 1s Apply complete! Resources: 0 added, 0 changed, 1 destroyed. ``` </details> --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cmu-eats_eats: Modifying... [id=scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_cmu-eats_eats: Modifications complete after 0s [id=scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba] keycloak_group_memberships.projects: Modifications complete after 2s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 2 changed, 0 destroyed. ``` </details> --- ### 6. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.23s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ googleplay_app_iam.eats_stanleymw: Destroying... googleplay_app_iam.eats_stanleymw: Destruction complete after 3s googleplay_user.stanleymw: Destroying... [id=smwang2@andrew.cmu.edu] googleplay_user.stanleymw: Destruction complete after 0s Apply complete! Resources: 0 added, 0 changed, 2 destroyed. ``` </details> --- ### Apply Summary 6 projects, 6 successful, 0 failed, 0 errored
xboxbedrock deleted branch SH-main 2026-09-28 16:25:38 +00:00

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
xboxbedrock referenced this pull request from a commit 2026-09-28 16:25:40 +00:00
Sign in to join this conversation.
No description provided.