feat: add armaanss, bxiang, jdpeters, sudipi to cmu-study #454

Merged
xboxbedrock merged 3 commits from sync/add-members-cmu-study into main 2026-09-28 15:50:17 +00:00
Owner

Consolidates #443, #446, #447, #430. Tofu regen will be committed before planning.

Consolidates #443, #446, #447, #430. Tofu regen will be committed before planning.
feat: add armaanss, bxiang, jdpeters, sudipi to cmu-study
Some checks failed
Validate / Governance (pull_request_target) Failing after 3m10s
CI / check-1 (pull_request) Failing after 3m47s
CI / check (pull_request) Failing after 0s
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
477cd42cd5
xboxbedrock force-pushed sync/add-members-cmu-study from 477cd42cd5
Some checks failed
Validate / Governance (pull_request_target) Failing after 3m10s
CI / check-1 (pull_request) Failing after 3m47s
CI / check (pull_request) Failing after 0s
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
to 9784c402d3
Some checks failed
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Has been cancelled
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
2026-09-28 15:40:41 +00:00
Compare
chore: tofu gen
All checks were successful
CI / build (pull_request) Has been skipped
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 8 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 2 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 4 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 4 to add, 0 to change, 0 to destroy.
atlantis/plan: discord Plan: 4 to add, 0 to change, 0 to destroy.
atlantis/plan 5/5 projects planned successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 5/5 projects applied successfully.
CI / check-1 (pull_request) Successful in 6m52s
CI / check (pull_request) Successful in 0s
Validate / Governance (pull_request_target) Successful in 5m37s
1370c494fc

Ran Plan for 5 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: keycloak dir: tofu/keycloak workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: git dir: tofu/git workspace: default
  5. project: discord dir: tofu/discord workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-study_ch0_armaanss will be created
+ resource "null_resource" "slack_cmu-study_ch0_armaanss" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C09KN6M3P53"
          + "user"    = "U09EM73V9SN"
        }
    }

  # null_resource.slack_cmu-study_ch0_bxiang will be created
+ resource "null_resource" "slack_cmu-study_ch0_bxiang" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C09KN6M3P53"
          + "user"    = "U0BTFPB8W02"
        }
    }

  # null_resource.slack_cmu-study_ch0_jdpeters will be created
+ resource "null_resource" "slack_cmu-study_ch0_jdpeters" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C09KN6M3P53"
          + "user"    = "U0C11RN7YG5"
        }
    }

  # null_resource.slack_cmu-study_ch0_sudipi will be created
+ resource "null_resource" "slack_cmu-study_ch0_sudipi" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C09KN6M3P53"
          + "user"    = "U0C1EFVUBPW"
        }
    }

  # null_resource.slack_hub_armaanss will be created
+ resource "null_resource" "slack_hub_armaanss" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U09EM73V9SN"
        }
    }

  # null_resource.slack_hub_bxiang will be created
+ resource "null_resource" "slack_hub_bxiang" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0BTFPB8W02"
        }
    }

  # null_resource.slack_hub_jdpeters will be created
+ resource "null_resource" "slack_hub_jdpeters" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0C11RN7YG5"
        }
    }

  # null_resource.slack_hub_sudipi will be created
+ resource "null_resource" "slack_hub_sudipi" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0C1EFVUBPW"
        }
    }

Plan: 8 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 8 to add, 0 to change, 0 to destroy.


2. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.20s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-study_study will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-study_study" {
        id       = "scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a"
!       members  = [
+           "armaanss",
+           "bxiang",
+           "jdpeters",
+           "sudipi",
            # (8 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "armaanss",
+           "bxiang",
+           "jdpeters",
+           "sudipi",
            # (97 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 2 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 2 to change, 0 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.23s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_armaanss will be created
+ resource "google_cloud_identity_group_membership" "tech_armaanss" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "armaanss@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

  # google_cloud_identity_group_membership.tech_bxiang will be created
+ resource "google_cloud_identity_group_membership" "tech_bxiang" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "bxiang@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

  # google_cloud_identity_group_membership.tech_jdpeters will be created
+ resource "google_cloud_identity_group_membership" "tech_jdpeters" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "jdpeters@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

  # google_cloud_identity_group_membership.tech_sudipi will be created
+ resource "google_cloud_identity_group_membership" "tech_sudipi" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "sudipi@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 4 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 4 to add, 0 to change, 0 to destroy.


4. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.25s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # forgejo_team_member.cmu-study_armaanss will be created
+ resource "forgejo_team_member" "cmu-study_armaanss" {
+       team_id = 43
+       user    = "armaanss"
    }

  # forgejo_team_member.cmu-study_bxiang will be created
+ resource "forgejo_team_member" "cmu-study_bxiang" {
+       team_id = 43
+       user    = "bxiang"
    }

  # forgejo_team_member.cmu-study_jdpeters will be created
+ resource "forgejo_team_member" "cmu-study_jdpeters" {
+       team_id = 43
+       user    = "jdpeters"
    }

  # forgejo_team_member.cmu-study_sudipi will be created
+ resource "forgejo_team_member" "cmu-study_sudipi" {
+       team_id = 43
+       user    = "sudipi"
    }

Plan: 4 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 4 to add, 0 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.13s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.armaanss will be created
+ resource "discord_member_roles" "armaanss" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "639282991192866826"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1550659990266581074"
        }
    }

  # discord_member_roles.bxiang will be created
+ resource "discord_member_roles" "bxiang" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "1292663388538671209"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1550659990266581074"
        }
    }

  # discord_member_roles.jdpeters will be created
+ resource "discord_member_roles" "jdpeters" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "779012083609632830"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1550659990266581074"
        }
    }

  # discord_member_roles.sudipi will be created
+ resource "discord_member_roles" "sudipi" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "735941345117470870"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1550659990266581074"
        }
    }

Plan: 4 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 4 to add, 0 to change, 0 to destroy.


Plan Summary

5 projects, 5 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 5 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_cmu-study_ch0_armaanss will be created + resource "null_resource" "slack_cmu-study_ch0_armaanss" { + id = (known after apply) + triggers = { + "channel" = "C09KN6M3P53" + "user" = "U09EM73V9SN" } } # null_resource.slack_cmu-study_ch0_bxiang will be created + resource "null_resource" "slack_cmu-study_ch0_bxiang" { + id = (known after apply) + triggers = { + "channel" = "C09KN6M3P53" + "user" = "U0BTFPB8W02" } } # null_resource.slack_cmu-study_ch0_jdpeters will be created + resource "null_resource" "slack_cmu-study_ch0_jdpeters" { + id = (known after apply) + triggers = { + "channel" = "C09KN6M3P53" + "user" = "U0C11RN7YG5" } } # null_resource.slack_cmu-study_ch0_sudipi will be created + resource "null_resource" "slack_cmu-study_ch0_sudipi" { + id = (known after apply) + triggers = { + "channel" = "C09KN6M3P53" + "user" = "U0C1EFVUBPW" } } # null_resource.slack_hub_armaanss will be created + resource "null_resource" "slack_hub_armaanss" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U09EM73V9SN" } } # null_resource.slack_hub_bxiang will be created + resource "null_resource" "slack_hub_bxiang" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0BTFPB8W02" } } # null_resource.slack_hub_jdpeters will be created + resource "null_resource" "slack_hub_jdpeters" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0C11RN7YG5" } } # null_resource.slack_hub_sudipi will be created + resource "null_resource" "slack_hub_sudipi" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0C1EFVUBPW" } } Plan: 8 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 8 to add, 0 to change, 0 to destroy. --- ### 2. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.20s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-study_study will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-study_study" { id = "scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a" ! members = [ + "armaanss", + "bxiang", + "jdpeters", + "sudipi", # (8 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "armaanss", + "bxiang", + "jdpeters", + "sudipi", # (97 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 2 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 2 to change, 0 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.23s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_armaanss will be created + resource "google_cloud_identity_group_membership" "tech_armaanss" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "armaanss@andrew.cmu.edu" } + roles { + name = "MEMBER" } } # google_cloud_identity_group_membership.tech_bxiang will be created + resource "google_cloud_identity_group_membership" "tech_bxiang" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "bxiang@andrew.cmu.edu" } + roles { + name = "MEMBER" } } # google_cloud_identity_group_membership.tech_jdpeters will be created + resource "google_cloud_identity_group_membership" "tech_jdpeters" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "jdpeters@andrew.cmu.edu" } + roles { + name = "MEMBER" } } # google_cloud_identity_group_membership.tech_sudipi will be created + resource "google_cloud_identity_group_membership" "tech_sudipi" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "sudipi@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 4 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 4 to add, 0 to change, 0 to destroy. --- ### 4. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.25s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # forgejo_team_member.cmu-study_armaanss will be created + resource "forgejo_team_member" "cmu-study_armaanss" { + team_id = 43 + user = "armaanss" } # forgejo_team_member.cmu-study_bxiang will be created + resource "forgejo_team_member" "cmu-study_bxiang" { + team_id = 43 + user = "bxiang" } # forgejo_team_member.cmu-study_jdpeters will be created + resource "forgejo_team_member" "cmu-study_jdpeters" { + team_id = 43 + user = "jdpeters" } # forgejo_team_member.cmu-study_sudipi will be created + resource "forgejo_team_member" "cmu-study_sudipi" { + team_id = 43 + user = "sudipi" } Plan: 4 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 4 to add, 0 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.13s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.armaanss will be created + resource "discord_member_roles" "armaanss" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "639282991192866826" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1550659990266581074" } } # discord_member_roles.bxiang will be created + resource "discord_member_roles" "bxiang" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "1292663388538671209" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1550659990266581074" } } # discord_member_roles.jdpeters will be created + resource "discord_member_roles" "jdpeters" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "779012083609632830" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1550659990266581074" } } # discord_member_roles.sudipi will be created + resource "discord_member_roles" "sudipi" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "735941345117470870" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1550659990266581074" } } Plan: 4 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 4 to add, 0 to change, 0 to destroy. --- ### Plan Summary 5 projects, 5 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis apply

atlantis apply

Ran Apply for 5 projects:

  1. project: git dir: tofu/git workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: discord dir: tofu/discord workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default

1. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.12s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.cmu-study_jdpeters: Creating...
forgejo_team_member.cmu-study_bxiang: Creating...
forgejo_team_member.cmu-study_armaanss: Creating...
forgejo_team_member.cmu-study_sudipi: Creating...
forgejo_team_member.cmu-study_jdpeters: Creation complete after 0s
forgejo_team_member.cmu-study_bxiang: Creation complete after 0s
forgejo_team_member.cmu-study_sudipi: Creation complete after 0s
forgejo_team_member.cmu-study_armaanss: Creation complete after 0s

Apply complete! Resources: 4 added, 0 changed, 0 destroyed.

2. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.23s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_hub_sudipi: Creating...
null_resource.slack_cmu-study_ch0_jdpeters: Creating...
null_resource.slack_hub_sudipi: Provisioning with 'local-exec'...
null_resource.slack_hub_sudipi (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C1EFVUBPW"]
null_resource.slack_cmu-study_ch0_sudipi: Creating...
null_resource.slack_hub_armaanss: Creating...
null_resource.slack_cmu-study_ch0_armaanss: Creating...
null_resource.slack_hub_bxiang: Creating...
null_resource.slack_hub_armaanss: Provisioning with 'local-exec'...
null_resource.slack_hub_armaanss (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U09EM73V9SN"]
null_resource.slack_cmu-study_ch0_bxiang: Creating...
null_resource.slack_hub_bxiang: Provisioning with 'local-exec'...
null_resource.slack_hub_bxiang (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0BTFPB8W02"]
null_resource.slack_cmu-study_ch0_sudipi: Provisioning with 'local-exec'...
null_resource.slack_cmu-study_ch0_sudipi (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C09KN6M3P53 --user U0C1EFVUBPW"]
null_resource.slack_hub_jdpeters: Creating...
null_resource.slack_cmu-study_ch0_bxiang: Provisioning with 'local-exec'...
null_resource.slack_cmu-study_ch0_bxiang (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C09KN6M3P53 --user U0BTFPB8W02"]
null_resource.slack_hub_jdpeters: Provisioning with 'local-exec'...
null_resource.slack_hub_jdpeters (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C11RN7YG5"]
null_resource.slack_cmu-study_ch0_armaanss: Provisioning with 'local-exec'...
null_resource.slack_cmu-study_ch0_armaanss (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C09KN6M3P53 --user U09EM73V9SN"]
null_resource.slack_cmu-study_ch0_jdpeters: Provisioning with 'local-exec'...
null_resource.slack_cmu-study_ch0_jdpeters (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C09KN6M3P53 --user U0C11RN7YG5"]
null_resource.slack_hub_sudipi: Creation complete after 0s [id=8400163081741010150]
null_resource.slack_hub_armaanss: Creation complete after 0s [id=6203438846774494093]
null_resource.slack_cmu-study_ch0_jdpeters: Creation complete after 0s [id=8112109653380131948]
null_resource.slack_cmu-study_ch0_armaanss: Creation complete after 0s [id=5410115035413047386]
null_resource.slack_cmu-study_ch0_bxiang: Creation complete after 0s [id=3662744159739909849]
null_resource.slack_cmu-study_ch0_sudipi: Creation complete after 0s [id=3798388597006205060]
null_resource.slack_hub_bxiang: Creation complete after 0s [id=8871746193193948906]
null_resource.slack_hub_jdpeters: Creation complete after 0s [id=7421827637212212985]

Apply complete! Resources: 8 added, 0 changed, 0 destroyed.

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.17s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_bxiang: Creating...
google_cloud_identity_group_membership.tech_armaanss: Creating...
google_cloud_identity_group_membership.tech_sudipi: Creating...
google_cloud_identity_group_membership.tech_jdpeters: Creating...
google_cloud_identity_group_membership.tech_jdpeters: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/104195286105387762142]
google_cloud_identity_group_membership.tech_armaanss: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/115946035627561359960]
google_cloud_identity_group_membership.tech_sudipi: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/114115271889660352744]
google_cloud_identity_group_membership.tech_bxiang: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/114564658235872908231]

Apply complete! Resources: 4 added, 0 changed, 0 destroyed.

4. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.sudipi: Creating...
discord_member_roles.bxiang: Creating...
discord_member_roles.jdpeters: Creating...
discord_member_roles.armaanss: Creating...
discord_member_roles.jdpeters: Creation complete after 2s [id=1353118209024331948:779012083609632830]
discord_member_roles.bxiang: Creation complete after 3s [id=1353118209024331948:1292663388538671209]
discord_member_roles.sudipi: Creation complete after 4s [id=1353118209024331948:735941345117470870]
discord_member_roles.armaanss: Creation complete after 4s [id=1353118209024331948:639282991192866826]

Apply complete! Resources: 4 added, 0 changed, 0 destroyed.

5. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.19s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cmu-study_study: Modifying... [id=scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_cmu-study_study: Modifications complete after 7s [id=scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a]
keycloak_group_memberships.projects: Modifications complete after 8s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 2 changed, 0 destroyed.

Apply Summary

5 projects, 5 successful, 0 failed, 0 errored

Ran Apply for 5 projects: 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.12s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.cmu-study_jdpeters: Creating... forgejo_team_member.cmu-study_bxiang: Creating... forgejo_team_member.cmu-study_armaanss: Creating... forgejo_team_member.cmu-study_sudipi: Creating... forgejo_team_member.cmu-study_jdpeters: Creation complete after 0s forgejo_team_member.cmu-study_bxiang: Creation complete after 0s forgejo_team_member.cmu-study_sudipi: Creation complete after 0s forgejo_team_member.cmu-study_armaanss: Creation complete after 0s Apply complete! Resources: 4 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.23s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_hub_sudipi: Creating... null_resource.slack_cmu-study_ch0_jdpeters: Creating... null_resource.slack_hub_sudipi: Provisioning with 'local-exec'... null_resource.slack_hub_sudipi (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C1EFVUBPW"] null_resource.slack_cmu-study_ch0_sudipi: Creating... null_resource.slack_hub_armaanss: Creating... null_resource.slack_cmu-study_ch0_armaanss: Creating... null_resource.slack_hub_bxiang: Creating... null_resource.slack_hub_armaanss: Provisioning with 'local-exec'... null_resource.slack_hub_armaanss (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U09EM73V9SN"] null_resource.slack_cmu-study_ch0_bxiang: Creating... null_resource.slack_hub_bxiang: Provisioning with 'local-exec'... null_resource.slack_hub_bxiang (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0BTFPB8W02"] null_resource.slack_cmu-study_ch0_sudipi: Provisioning with 'local-exec'... null_resource.slack_cmu-study_ch0_sudipi (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C09KN6M3P53 --user U0C1EFVUBPW"] null_resource.slack_hub_jdpeters: Creating... null_resource.slack_cmu-study_ch0_bxiang: Provisioning with 'local-exec'... null_resource.slack_cmu-study_ch0_bxiang (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C09KN6M3P53 --user U0BTFPB8W02"] null_resource.slack_hub_jdpeters: Provisioning with 'local-exec'... null_resource.slack_hub_jdpeters (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C11RN7YG5"] null_resource.slack_cmu-study_ch0_armaanss: Provisioning with 'local-exec'... null_resource.slack_cmu-study_ch0_armaanss (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C09KN6M3P53 --user U09EM73V9SN"] null_resource.slack_cmu-study_ch0_jdpeters: Provisioning with 'local-exec'... null_resource.slack_cmu-study_ch0_jdpeters (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C09KN6M3P53 --user U0C11RN7YG5"] null_resource.slack_hub_sudipi: Creation complete after 0s [id=8400163081741010150] null_resource.slack_hub_armaanss: Creation complete after 0s [id=6203438846774494093] null_resource.slack_cmu-study_ch0_jdpeters: Creation complete after 0s [id=8112109653380131948] null_resource.slack_cmu-study_ch0_armaanss: Creation complete after 0s [id=5410115035413047386] null_resource.slack_cmu-study_ch0_bxiang: Creation complete after 0s [id=3662744159739909849] null_resource.slack_cmu-study_ch0_sudipi: Creation complete after 0s [id=3798388597006205060] null_resource.slack_hub_bxiang: Creation complete after 0s [id=8871746193193948906] null_resource.slack_hub_jdpeters: Creation complete after 0s [id=7421827637212212985] Apply complete! Resources: 8 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.17s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_bxiang: Creating... google_cloud_identity_group_membership.tech_armaanss: Creating... google_cloud_identity_group_membership.tech_sudipi: Creating... google_cloud_identity_group_membership.tech_jdpeters: Creating... google_cloud_identity_group_membership.tech_jdpeters: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/104195286105387762142] google_cloud_identity_group_membership.tech_armaanss: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/115946035627561359960] google_cloud_identity_group_membership.tech_sudipi: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/114115271889660352744] google_cloud_identity_group_membership.tech_bxiang: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/114564658235872908231] Apply complete! Resources: 4 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.sudipi: Creating... discord_member_roles.bxiang: Creating... discord_member_roles.jdpeters: Creating... discord_member_roles.armaanss: Creating... discord_member_roles.jdpeters: Creation complete after 2s [id=1353118209024331948:779012083609632830] discord_member_roles.bxiang: Creation complete after 3s [id=1353118209024331948:1292663388538671209] discord_member_roles.sudipi: Creation complete after 4s [id=1353118209024331948:735941345117470870] discord_member_roles.armaanss: Creation complete after 4s [id=1353118209024331948:639282991192866826] Apply complete! Resources: 4 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.19s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cmu-study_study: Modifying... [id=scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_cmu-study_study: Modifications complete after 7s [id=scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a] keycloak_group_memberships.projects: Modifications complete after 8s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 2 changed, 0 destroyed. ``` </details> --- ### Apply Summary 5 projects, 5 successful, 0 failed, 0 errored
xboxbedrock deleted branch sync/add-members-cmu-study 2026-09-28 15:50:17 +00:00

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
Sign in to join this conversation.
No description provided.