feat: add locho to housing #429

Merged
thesuperRL merged 2 commits from sync/add-locho-housing into main 2026-09-27 18:16:06 +00:00
Owner

Clean re-submission superseding #372. Data-only change from main; tofu regen will be committed by CI.

Clean re-submission superseding #372. Data-only change from main; tofu regen will be committed by CI.
feat: add locho to housing
All checks were successful
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
Validate / Governance (pull_request_target) Successful in 2m57s
CI / check-1 (pull_request) Successful in 4m4s
CI / check (pull_request) Successful in 0s
CI / build (pull_request) Has been skipped
kennel/build skipped: no services or static sites defined
atlantis/plan 0/0 projects planned successfully.
atlantis/apply 0/0 projects applied successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
c41ab00072
Author
Owner

atlantis plan

atlantis plan

Ran Plan for 0 projects:

Ran Plan for 0 projects:
Author
Owner

atlantis plan

atlantis plan

Ran Plan for 0 projects:

Ran Plan for 0 projects:
Author
Owner

atlantis unlock

atlantis unlock

All Atlantis locks for this PR have been unlocked and plans discarded

All Atlantis locks for this PR have been unlocked and plans discarded
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 1 to add, 0 to change, 1 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 7 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 2 to add, 0 to change, 1 to destroy.
atlantis/plan: discord Plan: 1 to add, 1 to change, 0 to destroy.
atlantis/plan 5/5 projects planned successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 5/5 projects applied successfully.
76dc81a492

Ran Plan for 5 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: keycloak dir: tofu/keycloak workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: git dir: tofu/git workspace: default
  5. project: discord dir: tofu/discord workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on shared package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.28s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_cbp_ch0_lwahi will be destroyed
  # (because null_resource.slack_cbp_ch0_lwahi is not in configuration)
- resource "null_resource" "slack_cbp_ch0_lwahi" {
-       id       = "2277244147921481181" -> null
-       triggers = {
          - "channel" = "C08S56CJ8R2"
          - "user"    = "U0C18F0GT7G"
        } -> null
    }

  # null_resource.slack_hub_locho will be created
+ resource "null_resource" "slack_hub_locho" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0C12EHAGPP"
        }
    }

Plan: 1 to add, 0 to change, 1 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 1 to add, 0 to change, 1 to destroy.


2. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.29s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cbp_bus-sign will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_bus-sign" {
        id       = "scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b"
!       members  = [
-           "lwahi",
            # (19 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_dalmatian will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_dalmatian" {
        id       = "scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153"
!       members  = [
-           "lwahi",
            # (19 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_discord-verify will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_discord-verify" {
        id       = "scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745"
!       members  = [
-           "lwahi",
            # (19 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_groupme-mirror will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_groupme-mirror" {
        id       = "scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39"
!       members  = [
-           "lwahi",
            # (19 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-housing_housing" {
        id       = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d"
!       members  = [
+           "lxh",
            # (14 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_tartan-vote will be updated in-place
! resource "keycloak_group_memberships" "project_tartan-vote" {
        id       = "scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c"
!       members  = [
+           "lwahi",
            # (10 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "lxh",
            # (95 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 7 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 7 to change, 0 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.25s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_locho will be created
+ resource "google_cloud_identity_group_membership" "tech_locho" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "lxh@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


4. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.25s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # forgejo_team_member.cbp_lwahi will be destroyed
  # (because forgejo_team_member.cbp_lwahi is not in configuration)
- resource "forgejo_team_member" "cbp_lwahi" {
-       team_id = 3 -> null
-       user    = "lwahi" -> null
    }

  # forgejo_team_member.cmu-housing_locho will be created
+ resource "forgejo_team_member" "cmu-housing_locho" {
+       team_id = 2
+       user    = "locho"
    }

  # forgejo_team_member.tartan-vote_lwahi will be created
+ resource "forgejo_team_member" "tartan-vote_lwahi" {
+       team_id = 6
+       user    = "lwahi"
    }

Plan: 2 to add, 0 to change, 1 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 0 to change, 1 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.30s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # discord_member_roles.locho will be created
+ resource "discord_member_roles" "locho" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "582666638550368277"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1513653499957874871"
        }
    }

  # discord_member_roles.lwahi will be updated in-place
! resource "discord_member_roles" "lwahi" {
        id        = "1353118209024331948:738509196487688262"
        # (2 unchanged attributes hidden)

-       role {
-           has_role = true -> null
-           role_id  = "1494773174334783499" -> null
        }
+       role {
+           has_role = true
+           role_id  = "1508601984255524885"
        }

        # (1 unchanged block hidden)
    }

Plan: 1 to add, 1 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 1 to change, 0 to destroy.


Plan Summary

5 projects, 5 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 5 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on shared package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.28s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # null_resource.slack_cbp_ch0_lwahi will be destroyed # (because null_resource.slack_cbp_ch0_lwahi is not in configuration) - resource "null_resource" "slack_cbp_ch0_lwahi" { - id = "2277244147921481181" -> null - triggers = { - "channel" = "C08S56CJ8R2" - "user" = "U0C18F0GT7G" } -> null } # null_resource.slack_hub_locho will be created + resource "null_resource" "slack_hub_locho" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0C12EHAGPP" } } Plan: 1 to add, 0 to change, 1 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 1 to add, 0 to change, 1 to destroy. --- ### 2. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.29s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cbp_bus-sign will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_bus-sign" { id = "scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b" ! members = [ - "lwahi", # (19 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_dalmatian will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_dalmatian" { id = "scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153" ! members = [ - "lwahi", # (19 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_discord-verify will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_discord-verify" { id = "scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745" ! members = [ - "lwahi", # (19 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_groupme-mirror will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_groupme-mirror" { id = "scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39" ! members = [ - "lwahi", # (19 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-housing_housing" { id = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d" ! members = [ + "lxh", # (14 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_tartan-vote will be updated in-place ! resource "keycloak_group_memberships" "project_tartan-vote" { id = "scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c" ! members = [ + "lwahi", # (10 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "lxh", # (95 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 7 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 7 to change, 0 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.25s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_locho will be created + resource "google_cloud_identity_group_membership" "tech_locho" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "lxh@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 4. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.25s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # forgejo_team_member.cbp_lwahi will be destroyed # (because forgejo_team_member.cbp_lwahi is not in configuration) - resource "forgejo_team_member" "cbp_lwahi" { - team_id = 3 -> null - user = "lwahi" -> null } # forgejo_team_member.cmu-housing_locho will be created + resource "forgejo_team_member" "cmu-housing_locho" { + team_id = 2 + user = "locho" } # forgejo_team_member.tartan-vote_lwahi will be created + resource "forgejo_team_member" "tartan-vote_lwahi" { + team_id = 6 + user = "lwahi" } Plan: 2 to add, 0 to change, 1 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 0 to change, 1 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.30s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # discord_member_roles.locho will be created + resource "discord_member_roles" "locho" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "582666638550368277" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1513653499957874871" } } # discord_member_roles.lwahi will be updated in-place ! resource "discord_member_roles" "lwahi" { id = "1353118209024331948:738509196487688262" # (2 unchanged attributes hidden) - role { - has_role = true -> null - role_id = "1494773174334783499" -> null } + role { + has_role = true + role_id = "1508601984255524885" } # (1 unchanged block hidden) } Plan: 1 to add, 1 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 1 to change, 0 to destroy. --- ### Plan Summary 5 projects, 5 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Author
Owner

atlantis apply

atlantis apply

Ran Apply for 5 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: git dir: tofu/git workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_cbp_ch0_lwahi: Destroying... [id=2277244147921481181]
null_resource.slack_cbp_ch0_lwahi: Destruction complete after 0s
null_resource.slack_hub_locho: Creating...
null_resource.slack_hub_locho: Provisioning with 'local-exec'...
null_resource.slack_hub_locho (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C12EHAGPP"]
null_resource.slack_hub_locho: Creation complete after 0s [id=4360225298301891996]

Apply complete! Resources: 1 added, 0 changed, 1 destroyed.

2. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.17s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.cbp_lwahi: Destroying...
forgejo_team_member.cmu-housing_locho: Creating...
forgejo_team_member.tartan-vote_lwahi: Creating...
forgejo_team_member.tartan-vote_lwahi: Creation complete after 0s
forgejo_team_member.cbp_lwahi: Destruction complete after 0s
forgejo_team_member.cmu-housing_locho: Creation complete after 0s

Apply complete! Resources: 2 added, 0 changed, 1 destroyed.

3. project: discord dir: tofu/discord workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.09s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.locho: Creating...
discord_member_roles.lwahi: Modifying... [id=1353118209024331948:738509196487688262]
discord_member_roles.lwahi: Modifications complete after 0s [id=1353118209024331948:738509196487688262]
discord_member_roles.locho: Creation complete after 1s [id=1353118209024331948:582666638550368277]

Apply complete! Resources: 1 added, 1 changed, 0 destroyed.

4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.17s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_locho: Creating...
google_cloud_identity_group_membership.tech_locho: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/111069990805475038275]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

5. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cbp_dalmatian: Modifying... [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153]
keycloak_group_memberships.project_cmu-housing_housing: Modifying... [id=scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d]
keycloak_group_memberships.project_cbp_discord-verify: Modifying... [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745]
keycloak_group_memberships.project_cbp_bus-sign: Modifying... [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b]
keycloak_group_memberships.project_tartan-vote: Modifying... [id=scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c]
keycloak_group_memberships.project_cbp_groupme-mirror: Modifying... [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_cbp_dalmatian: Modifications complete after 0s [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153]
keycloak_group_memberships.project_cbp_discord-verify: Modifications complete after 0s [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745]
keycloak_group_memberships.project_cbp_groupme-mirror: Modifications complete after 0s [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39]
keycloak_group_memberships.project_cbp_bus-sign: Modifications complete after 0s [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b]
keycloak_group_memberships.project_tartan-vote: Modifications complete after 2s [id=scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c]
keycloak_group_memberships.project_cmu-housing_housing: Modifications complete after 2s [id=scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d]
keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 7 changed, 0 destroyed.

Apply Summary

5 projects, 5 successful, 0 failed, 0 errored

Ran Apply for 5 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_cbp_ch0_lwahi: Destroying... [id=2277244147921481181] null_resource.slack_cbp_ch0_lwahi: Destruction complete after 0s null_resource.slack_hub_locho: Creating... null_resource.slack_hub_locho: Provisioning with 'local-exec'... null_resource.slack_hub_locho (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C12EHAGPP"] null_resource.slack_hub_locho: Creation complete after 0s [id=4360225298301891996] Apply complete! Resources: 1 added, 0 changed, 1 destroyed. ``` </details> --- ### 2. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.17s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.cbp_lwahi: Destroying... forgejo_team_member.cmu-housing_locho: Creating... forgejo_team_member.tartan-vote_lwahi: Creating... forgejo_team_member.tartan-vote_lwahi: Creation complete after 0s forgejo_team_member.cbp_lwahi: Destruction complete after 0s forgejo_team_member.cmu-housing_locho: Creation complete after 0s Apply complete! Resources: 2 added, 0 changed, 1 destroyed. ``` </details> --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.09s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.locho: Creating... discord_member_roles.lwahi: Modifying... [id=1353118209024331948:738509196487688262] discord_member_roles.lwahi: Modifications complete after 0s [id=1353118209024331948:738509196487688262] discord_member_roles.locho: Creation complete after 1s [id=1353118209024331948:582666638550368277] Apply complete! Resources: 1 added, 1 changed, 0 destroyed. ``` </details> --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.17s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_locho: Creating... google_cloud_identity_group_membership.tech_locho: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/111069990805475038275] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cbp_dalmatian: Modifying... [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153] keycloak_group_memberships.project_cmu-housing_housing: Modifying... [id=scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d] keycloak_group_memberships.project_cbp_discord-verify: Modifying... [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745] keycloak_group_memberships.project_cbp_bus-sign: Modifying... [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b] keycloak_group_memberships.project_tartan-vote: Modifying... [id=scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c] keycloak_group_memberships.project_cbp_groupme-mirror: Modifying... [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_cbp_dalmatian: Modifications complete after 0s [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153] keycloak_group_memberships.project_cbp_discord-verify: Modifications complete after 0s [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745] keycloak_group_memberships.project_cbp_groupme-mirror: Modifications complete after 0s [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39] keycloak_group_memberships.project_cbp_bus-sign: Modifications complete after 0s [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b] keycloak_group_memberships.project_tartan-vote: Modifications complete after 2s [id=scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c] keycloak_group_memberships.project_cmu-housing_housing: Modifications complete after 2s [id=scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d] keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 7 changed, 0 destroyed. ``` </details> --- ### Apply Summary 5 projects, 5 successful, 0 failed, 0 errored
thesuperRL referenced this pull request from a commit 2026-09-27 18:16:08 +00:00

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
Sign in to join this conversation.
No description provided.