feat: add ecwang to cmu maps #424

Open
ecwang wants to merge 3 commits from ecwang/governance:main into main
First-time contributor
No description provided.
feat: add ecwang to cmu maps
Some checks are pending
CI / check (pull_request) Blocked by required conditions
CI / check-1 (pull_request) Blocked by required conditions
CI / build (pull_request) Blocked by required conditions
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 3m14s
492c0ae293
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: discord Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan 6/6 projects failed to plan.
ee27db8a04

Ran Plan for 6 projects:

  1. project: discord dir: tofu/discord workspace: default
  2. project: google_groups dir: tofu/google_groups workspace: default
  3. project: slack dir: tofu/slack workspace: default
  4. project: google_play dir: tofu/google_play workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: keycloak dir: tofu/keycloak workspace: default

1. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #423. To continue, delete the lock from #423 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #417. To continue, delete the lock from #417 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #423. To continue, delete the lock from #423 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #423. To continue, delete the lock from #423 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #423. To continue, delete the lock from #423 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #423. To continue, delete the lock from #423 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #417. To continue, delete the lock from #417 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #423. To continue, delete the lock from #423 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #423. To continue, delete the lock from #423 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #423. To continue, delete the lock from #423 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis plan

atlantis plan
Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/424/default": 
Download attempt 1 failed, retrying...
Download attempt 2 failed, retrying...
Download attempt 3 failed, retrying...
Failed to download forgejo provider after 3 attempts

``` Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/424/default": Download attempt 1 failed, retrying... Download attempt 2 failed, retrying... Download attempt 3 failed, retrying... Failed to download forgejo provider after 3 attempts ```
Owner

atlantis plan

atlantis plan
Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/424/default": 
Download attempt 1 failed, retrying...
Download attempt 2 failed, retrying...
Download attempt 3 failed, retrying...
Failed to download forgejo provider after 3 attempts

``` Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/424/default": Download attempt 1 failed, retrying... Download attempt 2 failed, retrying... Download attempt 3 failed, retrying... Failed to download forgejo provider after 3 attempts ```
Owner

atlantis plan

atlantis plan
Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/424/default": 
Download attempt 1 failed, retrying...
Download attempt 2 failed, retrying...
Download attempt 3 failed, retrying...
Failed to download forgejo provider after 3 attempts

``` Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/424/default": Download attempt 1 failed, retrying... Download attempt 2 failed, retrying... Download attempt 3 failed, retrying... Failed to download forgejo provider after 3 attempts ```
Owner

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: git dir: tofu/git workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default
  6. project: slack dir: tofu/slack workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #403. To continue, delete the lock from #403 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #403. To continue, delete the lock from #403 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: discord Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan 6/6 projects failed to plan.
This pull request has changes conflicting with the target branch.
  • data/teams/tartan-vote.toml
  • tofu/keycloak/keycloak_memberships.tf.json
Some workflows are waiting to be reviewed.
View command line instructions

Manual merge helper

Use this merge commit message when completing the merge manually.

Checkout

From your project repository, check out a new branch and test the changes.
git fetch -u main:ecwang-main
git switch ecwang-main

Merge

Merge the changes and update on Forgejo.

Warning: The "Autodetect manual merge" setting is not enabled for this repository, you will have to mark this pull request as manually merged afterwards.

git switch main
git merge --no-ff ecwang-main
git switch ecwang-main
git rebase main
git switch main
git merge --ff-only ecwang-main
git switch ecwang-main
git rebase main
git switch main
git merge --no-ff ecwang-main
git switch main
git merge --squash ecwang-main
git switch main
git merge --ff-only ecwang-main
git switch main
git merge ecwang-main
git push origin main
Sign in to join this conversation.
No description provided.