feat: remove hannahl3 from courses #383

Open
etashjha wants to merge 2 commits from etashjha/governance:main into main
Member
No description provided.
feat: remove hannahl3 from courses
All checks were successful
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 2m57s
CI / check-1 (pull_request) Successful in 4m11s
CI / check (pull_request) Successful in 0s
684769852b
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: google_groups Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan 6/6 projects failed to plan.
321cb74081

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_groups dir: tofu/google_groups workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: git dir: tofu/git workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default
  6. project: google_play dir: tofu/google_play workspace: default

1. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis plan

atlantis plan
Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/383/default": 
Download attempt 1 failed, retrying...
Download attempt 2 failed, retrying...
Download attempt 3 failed, retrying...
Failed to download forgejo provider after 3 attempts

``` Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/383/default": Download attempt 1 failed, retrying... Download attempt 2 failed, retrying... Download attempt 3 failed, retrying... Failed to download forgejo provider after 3 attempts ```
Owner

atlantis plan

atlantis plan
Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/383/default": 
Download attempt 1 failed, retrying...
Download attempt 2 failed, retrying...
Download attempt 3 failed, retrying...
Failed to download forgejo provider after 3 attempts

``` Error: Pre-workflow hook failed: exit status 1: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/383/default": Download attempt 1 failed, retrying... Download attempt 2 failed, retrying... Download attempt 3 failed, retrying... Failed to download forgejo provider after 3 attempts ```
Owner

atlantis plan

atlantis plan
Error: Pre-workflow hook failed: exit status 2: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/383/default": 
Download attempt 1 failed, retrying...
Download attempt 2 failed, retrying...
Archive:  /tmp/forgejo-provider.zip
  inflating: /tmp/forgejo-provider/CHANGELOG.md  
  inflating: /tmp/forgejo-provider/LICENSE  
  inflating: /tmp/forgejo-provider/README.md  
  inflating: /tmp/forgejo-provider/terraform-provider-forgejo_v1.4.2-team-repository.2  
  error:  invalid compressed data to inflate

``` Error: Pre-workflow hook failed: exit status 2: running "sh -c bash .forgejo/scripts/install-forgejo-provider.sh" in "/var/lib/atlantis/repos/ScottyLabs/governance/383/default": Download attempt 1 failed, retrying... Download attempt 2 failed, retrying... Archive: /tmp/forgejo-provider.zip inflating: /tmp/forgejo-provider/CHANGELOG.md inflating: /tmp/forgejo-provider/LICENSE inflating: /tmp/forgejo-provider/README.md inflating: /tmp/forgejo-provider/terraform-provider-forgejo_v1.4.2-team-repository.2 error: invalid compressed data to inflate ```
Owner

atlantis plan

atlantis plan

Ran Plan for 6 projects:

  1. project: discord dir: tofu/discord workspace: default
  2. project: git dir: tofu/git workspace: default
  3. project: slack dir: tofu/slack workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: google_groups dir: tofu/google_groups workspace: default
  6. project: google_play dir: tofu/google_play workspace: default

1. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` --- ### 1. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #418. To continue, delete the lock from #418 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #378. To continue, delete the lock from #378 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: google_groups Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan 6/6 projects failed to plan.
This pull request has changes conflicting with the target branch.
  • tofu/discord/discord.tf.json
  • tofu/google_groups/google_groups.tf.json
  • tofu/google_play/google_play.tf.json
  • tofu/keycloak/keycloak_memberships.tf.json
  • tofu/slack/slack.tf.json
View command line instructions

Manual merge helper

Use this merge commit message when completing the merge manually.

Checkout

From your project repository, check out a new branch and test the changes.
git fetch -u main:etashjha-main
git switch etashjha-main

Merge

Merge the changes and update on Forgejo.

Warning: The "Autodetect manual merge" setting is not enabled for this repository, you will have to mark this pull request as manually merged afterwards.

git switch main
git merge --no-ff etashjha-main
git switch etashjha-main
git rebase main
git switch main
git merge --ff-only etashjha-main
git switch etashjha-main
git rebase main
git switch main
git merge --no-ff etashjha-main
git switch main
git merge --squash etashjha-main
git switch main
git merge --ff-only etashjha-main
git switch main
git merge etashjha-main
git push origin main
Sign in to join this conversation.
No description provided.