chore: cleanup and adding SLAI #376
No reviewers
Labels
No labels
needs/linking
needs/docs
needs/research
needs/upstream
type/bug
type/external
type/feature
type/refactor
urgency
high
urgency
immediate
urgency
low
urgency
medium
No milestone
No project
No assignees
3 participants
Notifications
Due date
No due date set.
Dependencies
No dependencies set
Reference
ScottyLabs/governance!376
Loading…
Reference in a new issue
No description provided.
Delete branch "krishsax/governance:main"
Deleting a branch is permanent. Although the deleted branch may continue to exist for a short time before it actually gets removed, it CANNOT be undone in most cases. Continue?
thanks :D
Ran Plan for 3 projects:
keycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:defaultopenbaodir:tofu/openbaoworkspace:default1. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
openbaodir:tofu/openbaoworkspace:defaultShow Output
Plan: 6 to add, 50 to change, 0 to destroy.
Plan Summary
3 projects, 1 with changes, 0 with no changes, 2 failed
Ran Plan for 3 projects:
keycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:defaultopenbaodir:tofu/openbaoworkspace:default1. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
openbaodir:tofu/openbaoworkspace:defaultShow Output
Plan: 6 to add, 50 to change, 0 to destroy.
Plan Summary
3 projects, 1 with changes, 0 with no changes, 2 failed
chore: cleanup and adding SLAI Repoto chore: cleanup and adding SLAI RFC repoRan Plan for 3 projects:
gitdir:tofu/gitworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultopenbaodir:tofu/openbaoworkspace:default1. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
openbaodir:tofu/openbaoworkspace:defaultShow Output
Plan: 6 to add, 50 to change, 0 to destroy.
Plan Summary
3 projects, 1 with changes, 0 with no changes, 2 failed
chore: cleanup and adding SLAI RFC repoto chore: cleanup and adding SLAIRan Plan for 3 projects:
gitdir:tofu/gitworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultopenbaodir:tofu/openbaoworkspace:default1. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
openbaodir:tofu/openbaoworkspace:defaultShow Output
Plan: 6 to add, 50 to change, 0 to destroy.
Plan Summary
3 projects, 1 with changes, 0 with no changes, 2 failed
Ran Plan for 3 projects:
keycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:defaultopenbaodir:tofu/openbaoworkspace:default1. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #373. To continue, delete the lock from #373 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
openbaodir:tofu/openbaoworkspace:defaultShow Output
Plan: 6 to add, 50 to change, 0 to destroy.
Plan Summary
3 projects, 1 with changes, 0 with no changes, 2 failed
atlantis plan
Ran Plan for 3 projects:
openbaodir:tofu/openbaoworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
openbaodir:tofu/openbaoworkspace:defaultShow Output
Plan: 6 to add, 50 to change, 0 to destroy.
2. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 4 to add, 0 to change, 0 to destroy.
3. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 13 to add, 1 to change, 0 to destroy.
Plan Summary
3 projects, 3 with changes, 0 with no changes, 0 failed
atlantis apply
Ran Apply for 3 projects:
openbaodir:tofu/openbaoworkspace:defaultgitdir:tofu/gitworkspace:defaultkeycloakdir:tofu/keycloakworkspace:default1. project:
openbaodir:tofu/openbaoworkspace:defaultShow Output
2. project:
gitdir:tofu/gitworkspace:defaultShow Output
3. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Apply Summary
3 projects, 3 successful, 0 failed, 0 errored
Locks and plans deleted for the projects and workspaces modified in this pull request:
tofu/gitworkspace:defaulttofu/keycloakworkspace:defaulttofu/openbaoworkspace:default