feat: add walterh to slai #366
No reviewers
Labels
No labels
needs/linking
needs/docs
needs/research
needs/upstream
type/bug
type/external
type/feature
type/refactor
urgency
high
urgency
immediate
urgency
low
urgency
medium
No milestone
No project
No assignees
5 participants
Notifications
Due date
No due date set.
Dependencies
No dependencies set
Reference
ScottyLabs/governance!366
Loading…
Reference in a new issue
No description provided.
Delete branch "walterh/governance:main"
Deleting a branch is permanent. Although the deleted branch may continue to exist for a short time before it actually gets removed, it CANNOT be undone in most cases. Continue?
Signed-off-by: Walter Hsieh walterh@andrew.cmu.edu
Ran Plan for 5 projects:
discorddir:tofu/discordworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultslackdir:tofu/slackworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #364. To continue, delete the lock from #364 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #364. To continue, delete the lock from #364 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #364. To continue, delete the lock from #364 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #364. To continue, delete the lock from #364 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 0 to destroy.
Plan Summary
5 projects, 1 with changes, 0 with no changes, 4 failed
atlantis plan
Ran Plan for 5 projects:
slackdir:tofu/slackworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 0 to destroy.
2. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 0 to add, 5 to change, 0 to destroy.
3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 0 to destroy.
4. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 0 to destroy.
5. project:
discorddir:tofu/discordworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 0 to destroy.
Plan Summary
5 projects, 5 with changes, 0 with no changes, 0 failed
atlantis unlock
All Atlantis locks for this PR have been unlocked and plans discarded
superseded by #422
Pull request closed