feat: add js1 to cbp #362

Merged
thesuperRL merged 2 commits from js1/governance:main into main 2026-09-23 19:48:53 +00:00
Member
No description provided.
feat: add js1 to cbp
All checks were successful
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 2m53s
CI / check-1 (pull_request) Successful in 4m13s
CI / build (pull_request) Has been skipped
CI / check (pull_request) Successful in 0s
f3242a37a4
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: discord Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan 5/5 projects failed to plan.
kennel/build skipped: no services or static sites defined
2ce7801683

Ran Plan for 5 projects:

  1. project: discord dir: tofu/discord workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: git dir: tofu/git workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default

1. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #361. To continue, delete the lock from #361 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #361. To continue, delete the lock from #361 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #361. To continue, delete the lock from #361 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #361. To continue, delete the lock from #361 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #361. To continue, delete the lock from #361 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

5 projects, 0 with changes, 0 with no changes, 5 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 5 projects: 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #361. To continue, delete the lock from #361 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #361. To continue, delete the lock from #361 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #361. To continue, delete the lock from #361 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #361. To continue, delete the lock from #361 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #361. To continue, delete the lock from #361 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 5 projects, 0 with changes, 0 with no changes, 5 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
tentype approved these changes 2026-09-22 20:11:07 +00:00
thesuperRL force-pushed main from 2ce7801683
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: discord Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan 5/5 projects failed to plan.
kennel/build skipped: no services or static sites defined
to a11fcd789b
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: slack Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 5 to change, 0 to destroy.
atlantis/plan: git Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/apply: slack Apply succeeded.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan 5/5 projects planned successfully.
atlantis/apply: git Apply succeeded.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 5/5 projects applied successfully.
2026-09-23 19:46:12 +00:00
Compare

Ran Plan for 5 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: keycloak dir: tofu/keycloak workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: git dir: tofu/git workspace: default
  5. project: discord dir: tofu/discord workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/362/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/362/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 4.55s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_cbp_ch0_js1 will be created
+ resource "null_resource" "slack_cbp_ch0_js1" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08S56CJ8R2"
          + "user"    = "U0C368L5HL3"
        }
    }

  # null_resource.slack_hub_js1 will be created
+ resource "null_resource" "slack_hub_js1" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0C368L5HL3"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 0 to destroy.


2. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 4.59s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cbp_bus-sign will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_bus-sign" {
        id       = "scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b"
!       members  = [
+           "jihosung",
            # (18 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_dalmatian will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_dalmatian" {
        id       = "scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153"
!       members  = [
+           "jihosung",
            # (18 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_discord-verify will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_discord-verify" {
        id       = "scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745"
!       members  = [
+           "jihosung",
            # (18 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_groupme-mirror will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_groupme-mirror" {
        id       = "scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39"
!       members  = [
+           "jihosung",
            # (18 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "jihosung",
            # (91 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 5 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 5 to change, 0 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 4.54s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_js1 will be created
+ resource "google_cloud_identity_group_membership" "tech_js1" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "jihosung@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


4. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 4.57s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # forgejo_team_member.cbp_js1 will be created
+ resource "forgejo_team_member" "cbp_js1" {
+       team_id = 3
+       user    = "jihosung"
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 1 to add, 0 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 4.53s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.js1 will be created
+ resource "discord_member_roles" "js1" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "551918649120915477"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1494773174334783499"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


Plan Summary

5 projects, 5 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 5 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/362/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/362/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 4.55s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_cbp_ch0_js1 will be created + resource "null_resource" "slack_cbp_ch0_js1" { + id = (known after apply) + triggers = { + "channel" = "C08S56CJ8R2" + "user" = "U0C368L5HL3" } } # null_resource.slack_hub_js1 will be created + resource "null_resource" "slack_hub_js1" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0C368L5HL3" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 2. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 4.59s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cbp_bus-sign will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_bus-sign" { id = "scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b" ! members = [ + "jihosung", # (18 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_dalmatian will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_dalmatian" { id = "scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153" ! members = [ + "jihosung", # (18 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_discord-verify will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_discord-verify" { id = "scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745" ! members = [ + "jihosung", # (18 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_groupme-mirror will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_groupme-mirror" { id = "scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39" ! members = [ + "jihosung", # (18 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "jihosung", # (91 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 5 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 5 to change, 0 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 4.54s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_js1 will be created + resource "google_cloud_identity_group_membership" "tech_js1" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "jihosung@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 4. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 4.57s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # forgejo_team_member.cbp_js1 will be created + resource "forgejo_team_member" "cbp_js1" { + team_id = 3 + user = "jihosung" } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 4.53s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.js1 will be created + resource "discord_member_roles" "js1" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "551918649120915477" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1494773174334783499" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### Plan Summary 5 projects, 5 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis apply

atlantis apply

Ran Apply for 5 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: git dir: tofu/git workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.10s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_cbp_ch0_js1: Creating...
null_resource.slack_hub_js1: Creating...
null_resource.slack_hub_js1: Provisioning with 'local-exec'...
null_resource.slack_hub_js1 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C368L5HL3"]
null_resource.slack_cbp_ch0_js1: Provisioning with 'local-exec'...
null_resource.slack_cbp_ch0_js1 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08S56CJ8R2 --user U0C368L5HL3"]
null_resource.slack_hub_js1: Creation complete after 0s [id=8479821236672714746]
null_resource.slack_cbp_ch0_js1: Creation complete after 1s [id=5292545810412975083]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

2. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.cbp_js1: Creating...
forgejo_team_member.cbp_js1: Creation complete after 0s

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

3. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.14s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.js1: Creating...
discord_member_roles.js1: Creation complete after 1s [id=1353118209024331948:551918649120915477]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.19s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_js1: Creating...
google_cloud_identity_group_membership.tech_js1: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/100011389605262730771]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

5. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.19s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cbp_discord-verify: Modifying... [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745]
keycloak_group_memberships.project_cbp_bus-sign: Modifying... [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b]
keycloak_group_memberships.project_cbp_groupme-mirror: Modifying... [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39]
keycloak_group_memberships.project_cbp_dalmatian: Modifying... [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_cbp_discord-verify: Modifications complete after 2s [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745]
keycloak_group_memberships.project_cbp_groupme-mirror: Modifications complete after 2s [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39]
keycloak_group_memberships.project_cbp_dalmatian: Modifications complete after 2s [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153]
keycloak_group_memberships.project_cbp_bus-sign: Modifications complete after 2s [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b]
keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 5 changed, 0 destroyed.

Apply Summary

5 projects, 5 successful, 0 failed, 0 errored

Ran Apply for 5 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.10s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_cbp_ch0_js1: Creating... null_resource.slack_hub_js1: Creating... null_resource.slack_hub_js1: Provisioning with 'local-exec'... null_resource.slack_hub_js1 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C368L5HL3"] null_resource.slack_cbp_ch0_js1: Provisioning with 'local-exec'... null_resource.slack_cbp_ch0_js1 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08S56CJ8R2 --user U0C368L5HL3"] null_resource.slack_hub_js1: Creation complete after 0s [id=8479821236672714746] null_resource.slack_cbp_ch0_js1: Creation complete after 1s [id=5292545810412975083] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.cbp_js1: Creating... forgejo_team_member.cbp_js1: Creation complete after 0s Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.14s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.js1: Creating... discord_member_roles.js1: Creation complete after 1s [id=1353118209024331948:551918649120915477] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.19s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_js1: Creating... google_cloud_identity_group_membership.tech_js1: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/100011389605262730771] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.19s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cbp_discord-verify: Modifying... [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745] keycloak_group_memberships.project_cbp_bus-sign: Modifying... [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b] keycloak_group_memberships.project_cbp_groupme-mirror: Modifying... [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39] keycloak_group_memberships.project_cbp_dalmatian: Modifying... [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_cbp_discord-verify: Modifications complete after 2s [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745] keycloak_group_memberships.project_cbp_groupme-mirror: Modifications complete after 2s [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39] keycloak_group_memberships.project_cbp_dalmatian: Modifications complete after 2s [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153] keycloak_group_memberships.project_cbp_bus-sign: Modifications complete after 2s [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b] keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 5 changed, 0 destroyed. ``` </details> --- ### Apply Summary 5 projects, 5 successful, 0 failed, 0 errored

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
thesuperRL referenced this pull request from a commit 2026-09-23 19:48:54 +00:00
brightz referenced this pull request from a commit 2026-09-26 22:03:28 +00:00
Sign in to join this conversation.
No description provided.