feat: add junwonp to devops #357

Merged
xboxbedrock merged 3 commits from add-devops-final into main 2026-09-26 20:52:04 +00:00
Owner

Consolidates #245 #285 #324 #351

Consolidates #245 #285 #324 #351
feat: add pzhou2, junwonp to devops
Some checks failed
CI / check-1 (pull_request) Successful in 4m0s
CI / check (pull_request) Successful in 0s
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Failing after 2m38s
e7ac991330
xboxbedrock force-pushed add-devops-final from e7ac991330
Some checks failed
CI / check-1 (pull_request) Successful in 4m0s
CI / check (pull_request) Successful in 0s
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Failing after 2m38s
to d5aa3f747e
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
CI / build (pull_request) Has been skipped
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Has been cancelled
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
2026-09-26 20:35:38 +00:00
Compare
feat: re-scope to junwon
All checks were successful
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 2m56s
CI / check-1 (pull_request) Successful in 4m13s
CI / check (pull_request) Successful in 0s
6cef6c347e
xboxbedrock changed title from feat: add pzhou2, junwonp to devops to feat: add junwonp to devops 2026-09-26 20:38:02 +00:00
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 3 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 5 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 1 to add, 2 to change, 0 to destroy.
atlantis/plan 5/5 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply 5/5 projects applied successfully.
5e8653266f

Ran Plan for 5 projects:

  1. project: keycloak dir: tofu/keycloak workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: git dir: tofu/git workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: discord dir: tofu/discord workspace: default

1. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #388. To continue, delete the lock from #388 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #388. To continue, delete the lock from #388 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #388. To continue, delete the lock from #388 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #388. To continue, delete the lock from #388 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #388. To continue, delete the lock from #388 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

5 projects, 0 with changes, 0 with no changes, 5 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 5 projects: 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` --- ### 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #388. To continue, delete the lock from #388 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #388. To continue, delete the lock from #388 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #388. To continue, delete the lock from #388 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #388. To continue, delete the lock from #388 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #388. To continue, delete the lock from #388 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 5 projects, 0 with changes, 0 with no changes, 5 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis plan

atlantis plan

Ran Plan for 5 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: keycloak dir: tofu/keycloak workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: discord dir: tofu/discord workspace: default
  5. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 5.03s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_devops_ch0_junwonp will be created
+ resource "null_resource" "slack_devops_ch0_junwonp" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08K3Q77ZQF"
          + "user"    = "U0C31L4SXAA"
        }
    }

  # null_resource.slack_devops_ch1_junwonp will be created
+ resource "null_resource" "slack_devops_ch1_junwonp" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0BNV84NUTC"
          + "user"    = "U0C31L4SXAA"
        }
    }

  # null_resource.slack_hub_junwonp will be created
+ resource "null_resource" "slack_hub_junwonp" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0C31L4SXAA"
        }
    }

Plan: 3 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 3 to add, 0 to change, 0 to destroy.


2. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 5.03s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_devops will be updated in-place
! resource "keycloak_group_memberships" "project_devops" {
        id       = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7"
!       members  = [
+           "junwonp",
            # (9 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_changelog will be updated in-place
! resource "keycloak_group_memberships" "project_devops_changelog" {
        id       = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77"
!       members  = [
+           "junwonp",
            # (9 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_documentation will be updated in-place
! resource "keycloak_group_memberships" "project_devops_documentation" {
        id       = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f"
!       members  = [
+           "junwonp",
            # (9 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_devops_link-shortener will be updated in-place
! resource "keycloak_group_memberships" "project_devops_link-shortener" {
        id       = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935"
!       members  = [
+           "junwonp",
            # (9 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "junwonp",
            # (93 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 5 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 5 to change, 0 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 5.05s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_junwonp will be created
+ resource "google_cloud_identity_group_membership" "tech_junwonp" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "junwonp@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


4. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on shared package cache
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/357/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/357/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 5.00s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.junwonp will be created
+ resource "discord_member_roles" "junwonp" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "662802573467844629"

+       role {
+           has_role = true
+           role_id  = "1488787087598813215"
        }
+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 5.02s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # forgejo_team_member.devops_junwonp will be created
+ resource "forgejo_team_member" "devops_junwonp" {
+       team_id = 4
+       user    = "junwonp"
    }

  # github_repository.slai_cmugpt_agent will be updated in-place
! resource "github_repository" "slai_cmugpt_agent" {
        id                                      = "cmugpt-agent"
        name                                    = "cmugpt-agent"
        # (39 unchanged attributes hidden)

-       template {
-           include_all_branches = false -> null
-           owner                = "ScottyLabs" -> null
-           repository           = "python-template" -> null
        }

        # (1 unchanged block hidden)
    }

  # github_repository.slai_cmugpt_surface will be updated in-place
! resource "github_repository" "slai_cmugpt_surface" {
        id                                      = "cmugpt-surface"
        name                                    = "cmugpt-surface"
        # (39 unchanged attributes hidden)

-       template {
-           include_all_branches = false -> null
-           owner                = "scottylabs-labrador" -> null
-           repository           = "ScottyStack" -> null
        }

        # (1 unchanged block hidden)
    }

Plan: 1 to add, 2 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 1 to add, 2 to change, 0 to destroy.


Plan Summary

5 projects, 5 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 5 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 5.03s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_devops_ch0_junwonp will be created + resource "null_resource" "slack_devops_ch0_junwonp" { + id = (known after apply) + triggers = { + "channel" = "C08K3Q77ZQF" + "user" = "U0C31L4SXAA" } } # null_resource.slack_devops_ch1_junwonp will be created + resource "null_resource" "slack_devops_ch1_junwonp" { + id = (known after apply) + triggers = { + "channel" = "C0BNV84NUTC" + "user" = "U0C31L4SXAA" } } # null_resource.slack_hub_junwonp will be created + resource "null_resource" "slack_hub_junwonp" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0C31L4SXAA" } } Plan: 3 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 3 to add, 0 to change, 0 to destroy. --- ### 2. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 5.03s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_devops will be updated in-place ! resource "keycloak_group_memberships" "project_devops" { id = "scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7" ! members = [ + "junwonp", # (9 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_changelog will be updated in-place ! resource "keycloak_group_memberships" "project_devops_changelog" { id = "scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77" ! members = [ + "junwonp", # (9 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_documentation will be updated in-place ! resource "keycloak_group_memberships" "project_devops_documentation" { id = "scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f" ! members = [ + "junwonp", # (9 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_devops_link-shortener will be updated in-place ! resource "keycloak_group_memberships" "project_devops_link-shortener" { id = "scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935" ! members = [ + "junwonp", # (9 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "junwonp", # (93 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 5 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 5 to change, 0 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 5.05s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_junwonp will be created + resource "google_cloud_identity_group_membership" "tech_junwonp" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "junwonp@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 4. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on shared package cache Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/357/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/357/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 5.00s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.junwonp will be created + resource "discord_member_roles" "junwonp" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "662802573467844629" + role { + has_role = true + role_id = "1488787087598813215" } + role { + has_role = true + role_id = "1488787088118644848" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 5.02s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # forgejo_team_member.devops_junwonp will be created + resource "forgejo_team_member" "devops_junwonp" { + team_id = 4 + user = "junwonp" } # github_repository.slai_cmugpt_agent will be updated in-place ! resource "github_repository" "slai_cmugpt_agent" { id = "cmugpt-agent" name = "cmugpt-agent" # (39 unchanged attributes hidden) - template { - include_all_branches = false -> null - owner = "ScottyLabs" -> null - repository = "python-template" -> null } # (1 unchanged block hidden) } # github_repository.slai_cmugpt_surface will be updated in-place ! resource "github_repository" "slai_cmugpt_surface" { id = "cmugpt-surface" name = "cmugpt-surface" # (39 unchanged attributes hidden) - template { - include_all_branches = false -> null - owner = "scottylabs-labrador" -> null - repository = "ScottyStack" -> null } # (1 unchanged block hidden) } Plan: 1 to add, 2 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 1 to add, 2 to change, 0 to destroy. --- ### Plan Summary 5 projects, 5 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis apply

atlantis apply

Ran Apply for 5 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on shared package cache
    Finished `release` profile [optimized] target(s) in 0.20s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_devops_ch1_junwonp: Creating...
null_resource.slack_devops_ch0_junwonp: Creating...
null_resource.slack_hub_junwonp: Creating...
null_resource.slack_devops_ch1_junwonp: Provisioning with 'local-exec'...
null_resource.slack_devops_ch1_junwonp (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0BNV84NUTC --user U0C31L4SXAA"]
null_resource.slack_devops_ch0_junwonp: Provisioning with 'local-exec'...
null_resource.slack_devops_ch0_junwonp (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08K3Q77ZQF --user U0C31L4SXAA"]
null_resource.slack_hub_junwonp: Provisioning with 'local-exec'...
null_resource.slack_hub_junwonp (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C31L4SXAA"]
null_resource.slack_hub_junwonp: Creation complete after 0s [id=7012973114601905831]
null_resource.slack_devops_ch1_junwonp: Creation complete after 0s [id=9162332391230659035]
null_resource.slack_devops_ch0_junwonp: Creation complete after 0s [id=8978198533764068893]

Apply complete! Resources: 3 added, 0 changed, 0 destroyed.

2. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on shared package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.23s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.junwonp: Creating...
discord_member_roles.junwonp: Creation complete after 1s [id=1353118209024331948:662802573467844629]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.25s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_junwonp: Creating...
google_cloud_identity_group_membership.tech_junwonp: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/113786424116030298330]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.21s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_devops: Modifying... [id=scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7]
keycloak_group_memberships.project_devops_documentation: Modifying... [id=scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f]
keycloak_group_memberships.project_devops_changelog: Modifying... [id=scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77]
keycloak_group_memberships.project_devops_link-shortener: Modifying... [id=scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_devops: Modifications complete after 3s [id=scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7]
keycloak_group_memberships.project_devops_documentation: Modifications complete after 3s [id=scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f]
keycloak_group_memberships.project_devops_changelog: Modifications complete after 3s [id=scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77]
keycloak_group_memberships.project_devops_link-shortener: Modifications complete after 2s [id=scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935]
keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 5 changed, 0 destroyed.

5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.devops_junwonp: Creating...
forgejo_team_member.devops_junwonp: Creation complete after 0s
github_repository.slai_cmugpt_agent: Modifying... [id=cmugpt-agent]
github_repository.slai_cmugpt_surface: Modifying... [id=cmugpt-surface]
github_repository.slai_cmugpt_agent: Modifications complete after 4s [id=cmugpt-agent]
github_repository.slai_cmugpt_surface: Modifications complete after 4s [id=cmugpt-surface]

Apply complete! Resources: 1 added, 2 changed, 0 destroyed.

Apply Summary

5 projects, 5 successful, 0 failed, 0 errored

Ran Apply for 5 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on shared package cache Finished `release` profile [optimized] target(s) in 0.20s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_devops_ch1_junwonp: Creating... null_resource.slack_devops_ch0_junwonp: Creating... null_resource.slack_hub_junwonp: Creating... null_resource.slack_devops_ch1_junwonp: Provisioning with 'local-exec'... null_resource.slack_devops_ch1_junwonp (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0BNV84NUTC --user U0C31L4SXAA"] null_resource.slack_devops_ch0_junwonp: Provisioning with 'local-exec'... null_resource.slack_devops_ch0_junwonp (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08K3Q77ZQF --user U0C31L4SXAA"] null_resource.slack_hub_junwonp: Provisioning with 'local-exec'... null_resource.slack_hub_junwonp (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C31L4SXAA"] null_resource.slack_hub_junwonp: Creation complete after 0s [id=7012973114601905831] null_resource.slack_devops_ch1_junwonp: Creation complete after 0s [id=9162332391230659035] null_resource.slack_devops_ch0_junwonp: Creation complete after 0s [id=8978198533764068893] Apply complete! Resources: 3 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on shared package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.23s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.junwonp: Creating... discord_member_roles.junwonp: Creation complete after 1s [id=1353118209024331948:662802573467844629] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.25s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_junwonp: Creating... google_cloud_identity_group_membership.tech_junwonp: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/113786424116030298330] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.21s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_devops: Modifying... [id=scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7] keycloak_group_memberships.project_devops_documentation: Modifying... [id=scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f] keycloak_group_memberships.project_devops_changelog: Modifying... [id=scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77] keycloak_group_memberships.project_devops_link-shortener: Modifying... [id=scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_devops: Modifications complete after 3s [id=scottylabs/group-memberships/25f7b55f-5ad9-497f-9f38-62f9113a5fb7] keycloak_group_memberships.project_devops_documentation: Modifications complete after 3s [id=scottylabs/group-memberships/3c072d0d-78f5-40d2-9b28-06eb1563eb2f] keycloak_group_memberships.project_devops_changelog: Modifications complete after 3s [id=scottylabs/group-memberships/a854d323-1d2d-4eaa-a249-1da2827b0d77] keycloak_group_memberships.project_devops_link-shortener: Modifications complete after 2s [id=scottylabs/group-memberships/3cc5ba5b-6117-4cf3-8fb9-891c3d31d935] keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 5 changed, 0 destroyed. ``` </details> --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.devops_junwonp: Creating... forgejo_team_member.devops_junwonp: Creation complete after 0s github_repository.slai_cmugpt_agent: Modifying... [id=cmugpt-agent] github_repository.slai_cmugpt_surface: Modifying... [id=cmugpt-surface] github_repository.slai_cmugpt_agent: Modifications complete after 4s [id=cmugpt-agent] github_repository.slai_cmugpt_surface: Modifications complete after 4s [id=cmugpt-surface] Apply complete! Resources: 1 added, 2 changed, 0 destroyed. ``` </details> --- ### Apply Summary 5 projects, 5 successful, 0 failed, 0 errored
xboxbedrock deleted branch add-devops-final 2026-09-26 20:52:04 +00:00

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
brightz referenced this pull request from a commit 2026-09-26 22:03:28 +00:00
Sign in to join this conversation.
No description provided.