feat: add jazang, ethanli to terrier #343

Merged
anish merged 2 commits from add-terrier-unapproved into main 2026-09-22 10:25:41 +00:00
Owner

Consolidates unapproved: #229 #327

Consolidates unapproved: #229 #327
feat: add jazang, ethanli to terrier (unapproved)
All checks were successful
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 20m5s
CI / check-1 (pull_request) Successful in 18m15s
CI / check (pull_request) Successful in 0s
563b22c35d
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 4 to add, 0 to change, 2 to destroy.
atlantis/plan: google_play Plan: 4 to add, 0 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 2 to add, 0 to change, 1 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 3 to change, 0 to destroy.
atlantis/plan: discord Plan: 2 to add, 0 to change, 1 to destroy.
atlantis/plan: git Plan: 4 to add, 0 to change, 1 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
f7add78e66

Ran Plan for 6 projects:

  1. project: google_groups dir: tofu/google_groups workspace: default
  2. project: git dir: tofu/git workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: slack dir: tofu/slack workspace: default
  5. project: google_play dir: tofu/google_play workspace: default
  6. project: keycloak dir: tofu/keycloak workspace: default

1. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #336. To continue, delete the lock from #336 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #336. To continue, delete the lock from #336 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #336. To continue, delete the lock from #336 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #336. To continue, delete the lock from #336 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #336. To continue, delete the lock from #336 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #336. To continue, delete the lock from #336 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #336. To continue, delete the lock from #336 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #336. To continue, delete the lock from #336 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #336. To continue, delete the lock from #336 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #336. To continue, delete the lock from #336 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #336. To continue, delete the lock from #336 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #336. To continue, delete the lock from #336 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.16s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-study_ch0_armaanss will be destroyed
  # (because null_resource.slack_cmu-study_ch0_armaanss is not in configuration)
- resource "null_resource" "slack_cmu-study_ch0_armaanss" {
-       id       = "6301904129149314234" -> null
-       triggers = {
          - "channel" = "C09KN6M3P53"
          - "user"    = "U09EM73V9SN"
        } -> null
    }

  # null_resource.slack_hub_armaanss will be destroyed
  # (because null_resource.slack_hub_armaanss is not in configuration)
- resource "null_resource" "slack_hub_armaanss" {
-       id       = "7873933317574871801" -> null
-       triggers = {
          - "channel" = "C096TM8EMS8"
          - "user"    = "U09EM73V9SN"
        } -> null
    }

  # null_resource.slack_hub_ethanli will be created
+ resource "null_resource" "slack_hub_ethanli" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U09F9AJBQAD"
        }
    }

  # null_resource.slack_hub_jazang will be created
+ resource "null_resource" "slack_hub_jazang" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0AAAKB971A"
        }
    }

  # null_resource.slack_terrier_ch0_ethanli will be created
+ resource "null_resource" "slack_terrier_ch0_ethanli" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08AAADPCSE"
          + "user"    = "U09F9AJBQAD"
        }
    }

  # null_resource.slack_terrier_ch0_jazang will be created
+ resource "null_resource" "slack_terrier_ch0_jazang" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08AAADPCSE"
          + "user"    = "U0AAAKB971A"
        }
    }

Plan: 4 to add, 0 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 4 to add, 0 to change, 2 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.15s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # googleplay_app_iam.terrier_ethanli will be created
+ resource "googleplay_app_iam" "terrier_ethanli" {
+       app_id               = "build.terrier.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "ethanli@andrew.cmu.edu"
    }

  # googleplay_app_iam.terrier_jazang will be created
+ resource "googleplay_app_iam" "terrier_jazang" {
+       app_id               = "build.terrier.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "jazang@andrew.cmu.edu"
    }

  # googleplay_user.ethanli will be created
+ resource "googleplay_user" "ethanli" {
+       email                = "ethanli@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

  # googleplay_user.jazang will be created
+ resource "googleplay_user" "jazang" {
+       email                = "jazang@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

Plan: 4 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 4 to add, 0 to change, 0 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.25s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_armaanss will be destroyed
  # (because google_cloud_identity_group_membership.tech_armaanss is not in configuration)
- resource "google_cloud_identity_group_membership" "tech_armaanss" {
-       create_time = "2026-09-22T10:01:22.658259Z" -> null
-       group       = "groups/04anzqyu1bqbv79" -> null
-       id          = "groups/04anzqyu1bqbv79/memberships/115946035627561359960" -> null
-       name        = "groups/04anzqyu1bqbv79/memberships/115946035627561359960" -> null
-       type        = "USER" -> null
-       update_time = "2026-09-22T10:01:22.658259Z" -> null

-       preferred_member_key {
-           id = "armaanss@andrew.cmu.edu" -> null
        }

-       roles {
-           name = "MEMBER" -> null
        }
    }

  # google_cloud_identity_group_membership.tech_ethanli will be created
+ resource "google_cloud_identity_group_membership" "tech_ethanli" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "ethanli@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

  # google_cloud_identity_group_membership.tech_jazang will be created
+ resource "google_cloud_identity_group_membership" "tech_jazang" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "jazang@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 2 to add, 0 to change, 1 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 2 to add, 0 to change, 1 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.24s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-study_study will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-study_study" {
        id       = "scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a"
!       members  = [
-           "armaanss",
            # (8 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_terrier will be updated in-place
! resource "keycloak_group_memberships" "project_terrier" {
        id       = "scottylabs/group-memberships/678c7fa9-35cc-4bef-9a06-87cb6aa36238"
!       members  = [
+           "ethanli",
+           "jazang",
            # (6 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
-           "armaanss",
+           "ethanli",
+           "jazang",
            # (75 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 3 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 3 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.23s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # discord_member_roles.armaanss will be destroyed
  # (because discord_member_roles.armaanss is not in configuration)
- resource "discord_member_roles" "armaanss" {
-       id        = "1353118209024331948:639282991192866826" -> null
-       server_id = "1353118209024331948" -> null
-       user_id   = "639282991192866826" -> null

-       role {
-           has_role = true -> null
-           role_id  = "1488787088118644848" -> null
        }
-       role {
-           has_role = true -> null
-           role_id  = "1550659990266581074" -> null
        }
    }

  # discord_member_roles.ethanli will be created
+ resource "discord_member_roles" "ethanli" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "770051855857221693"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1517227361790201857"
        }
    }

  # discord_member_roles.jazang will be created
+ resource "discord_member_roles" "jazang" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "1412952264556941466"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1517227361790201857"
        }
    }

Plan: 2 to add, 0 to change, 1 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 2 to add, 0 to change, 1 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.17s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
- destroy

OpenTofu will perform the following actions:

  # forgejo_team_member.cmu-study_armaanss will be destroyed
  # (because forgejo_team_member.cmu-study_armaanss is not in configuration)
- resource "forgejo_team_member" "cmu-study_armaanss" {
-       team_id = 43 -> null
-       user    = "armaanss" -> null
    }

  # forgejo_team_member.terrier_ethanli will be created
+ resource "forgejo_team_member" "terrier_ethanli" {
+       team_id = 10
+       user    = "ethanli"
    }

  # forgejo_team_member.terrier_jazang will be created
+ resource "forgejo_team_member" "terrier_jazang" {
+       team_id = 10
+       user    = "jazang"
    }

  # github_team_membership.terrier_ethanli will be created
+ resource "github_team_membership" "terrier_ethanli" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "12028577"
+       username = "elsaber505"
    }

  # github_team_membership.terrier_jazang will be created
+ resource "github_team_membership" "terrier_jazang" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "12028577"
+       username = "jacqzang"
    }

Plan: 4 to add, 0 to change, 1 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 4 to add, 0 to change, 1 to destroy.


Plan Summary

6 projects, 6 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.16s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # null_resource.slack_cmu-study_ch0_armaanss will be destroyed # (because null_resource.slack_cmu-study_ch0_armaanss is not in configuration) - resource "null_resource" "slack_cmu-study_ch0_armaanss" { - id = "6301904129149314234" -> null - triggers = { - "channel" = "C09KN6M3P53" - "user" = "U09EM73V9SN" } -> null } # null_resource.slack_hub_armaanss will be destroyed # (because null_resource.slack_hub_armaanss is not in configuration) - resource "null_resource" "slack_hub_armaanss" { - id = "7873933317574871801" -> null - triggers = { - "channel" = "C096TM8EMS8" - "user" = "U09EM73V9SN" } -> null } # null_resource.slack_hub_ethanli will be created + resource "null_resource" "slack_hub_ethanli" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U09F9AJBQAD" } } # null_resource.slack_hub_jazang will be created + resource "null_resource" "slack_hub_jazang" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0AAAKB971A" } } # null_resource.slack_terrier_ch0_ethanli will be created + resource "null_resource" "slack_terrier_ch0_ethanli" { + id = (known after apply) + triggers = { + "channel" = "C08AAADPCSE" + "user" = "U09F9AJBQAD" } } # null_resource.slack_terrier_ch0_jazang will be created + resource "null_resource" "slack_terrier_ch0_jazang" { + id = (known after apply) + triggers = { + "channel" = "C08AAADPCSE" + "user" = "U0AAAKB971A" } } Plan: 4 to add, 0 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 4 to add, 0 to change, 2 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.15s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # googleplay_app_iam.terrier_ethanli will be created + resource "googleplay_app_iam" "terrier_ethanli" { + app_id = "build.terrier.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "ethanli@andrew.cmu.edu" } # googleplay_app_iam.terrier_jazang will be created + resource "googleplay_app_iam" "terrier_jazang" { + app_id = "build.terrier.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "jazang@andrew.cmu.edu" } # googleplay_user.ethanli will be created + resource "googleplay_user" "ethanli" { + email = "ethanli@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } # googleplay_user.jazang will be created + resource "googleplay_user" "jazang" { + email = "jazang@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } Plan: 4 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 4 to add, 0 to change, 0 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.25s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_armaanss will be destroyed # (because google_cloud_identity_group_membership.tech_armaanss is not in configuration) - resource "google_cloud_identity_group_membership" "tech_armaanss" { - create_time = "2026-09-22T10:01:22.658259Z" -> null - group = "groups/04anzqyu1bqbv79" -> null - id = "groups/04anzqyu1bqbv79/memberships/115946035627561359960" -> null - name = "groups/04anzqyu1bqbv79/memberships/115946035627561359960" -> null - type = "USER" -> null - update_time = "2026-09-22T10:01:22.658259Z" -> null - preferred_member_key { - id = "armaanss@andrew.cmu.edu" -> null } - roles { - name = "MEMBER" -> null } } # google_cloud_identity_group_membership.tech_ethanli will be created + resource "google_cloud_identity_group_membership" "tech_ethanli" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "ethanli@andrew.cmu.edu" } + roles { + name = "MEMBER" } } # google_cloud_identity_group_membership.tech_jazang will be created + resource "google_cloud_identity_group_membership" "tech_jazang" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "jazang@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 2 to add, 0 to change, 1 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 2 to add, 0 to change, 1 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.24s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-study_study will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-study_study" { id = "scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a" ! members = [ - "armaanss", # (8 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_terrier will be updated in-place ! resource "keycloak_group_memberships" "project_terrier" { id = "scottylabs/group-memberships/678c7fa9-35cc-4bef-9a06-87cb6aa36238" ! members = [ + "ethanli", + "jazang", # (6 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ - "armaanss", + "ethanli", + "jazang", # (75 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 3 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 3 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.23s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # discord_member_roles.armaanss will be destroyed # (because discord_member_roles.armaanss is not in configuration) - resource "discord_member_roles" "armaanss" { - id = "1353118209024331948:639282991192866826" -> null - server_id = "1353118209024331948" -> null - user_id = "639282991192866826" -> null - role { - has_role = true -> null - role_id = "1488787088118644848" -> null } - role { - has_role = true -> null - role_id = "1550659990266581074" -> null } } # discord_member_roles.ethanli will be created + resource "discord_member_roles" "ethanli" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "770051855857221693" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1517227361790201857" } } # discord_member_roles.jazang will be created + resource "discord_member_roles" "jazang" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "1412952264556941466" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1517227361790201857" } } Plan: 2 to add, 0 to change, 1 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 2 to add, 0 to change, 1 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.17s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create - destroy OpenTofu will perform the following actions: # forgejo_team_member.cmu-study_armaanss will be destroyed # (because forgejo_team_member.cmu-study_armaanss is not in configuration) - resource "forgejo_team_member" "cmu-study_armaanss" { - team_id = 43 -> null - user = "armaanss" -> null } # forgejo_team_member.terrier_ethanli will be created + resource "forgejo_team_member" "terrier_ethanli" { + team_id = 10 + user = "ethanli" } # forgejo_team_member.terrier_jazang will be created + resource "forgejo_team_member" "terrier_jazang" { + team_id = 10 + user = "jazang" } # github_team_membership.terrier_ethanli will be created + resource "github_team_membership" "terrier_ethanli" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "12028577" + username = "elsaber505" } # github_team_membership.terrier_jazang will be created + resource "github_team_membership" "terrier_jazang" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "12028577" + username = "jacqzang" } Plan: 4 to add, 0 to change, 1 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 4 to add, 0 to change, 1 to destroy. --- ### Plan Summary 6 projects, 6 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis apply

atlantis apply

Ran Apply for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_groups dir: tofu/google_groups workspace: default
  3. project: google_play dir: tofu/google_play workspace: default
  4. project: discord dir: tofu/discord workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: keycloak dir: tofu/keycloak workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.23s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_hub_armaanss: Destroying... [id=7873933317574871801]
null_resource.slack_cmu-study_ch0_armaanss: Destroying... [id=6301904129149314234]
null_resource.slack_hub_armaanss: Destruction complete after 0s
null_resource.slack_cmu-study_ch0_armaanss: Destruction complete after 0s
null_resource.slack_terrier_ch0_ethanli: Creating...
null_resource.slack_hub_ethanli: Creating...
null_resource.slack_terrier_ch0_ethanli: Provisioning with 'local-exec'...
null_resource.slack_terrier_ch0_ethanli (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08AAADPCSE --user U09F9AJBQAD"]
null_resource.slack_hub_jazang: Creating...
null_resource.slack_terrier_ch0_jazang: Creating...
null_resource.slack_hub_ethanli: Provisioning with 'local-exec'...
null_resource.slack_hub_ethanli (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U09F9AJBQAD"]
null_resource.slack_terrier_ch0_jazang: Provisioning with 'local-exec'...
null_resource.slack_terrier_ch0_jazang (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08AAADPCSE --user U0AAAKB971A"]
null_resource.slack_hub_jazang: Provisioning with 'local-exec'...
null_resource.slack_hub_jazang (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0AAAKB971A"]
null_resource.slack_terrier_ch0_ethanli: Creation complete after 0s [id=2370713295190598071]
null_resource.slack_hub_ethanli: Creation complete after 0s [id=8191639316496535190]
null_resource.slack_hub_jazang: Creation complete after 0s [id=8147252572916099479]
null_resource.slack_terrier_ch0_jazang: Creation complete after 0s [id=6186309242297098640]

Apply complete! Resources: 4 added, 0 changed, 2 destroyed.

2. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.19s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_armaanss: Destroying... [id=groups/04anzqyu1bqbv79/memberships/115946035627561359960]
google_cloud_identity_group_membership.tech_ethanli: Creating...
google_cloud_identity_group_membership.tech_jazang: Creating...
google_cloud_identity_group_membership.tech_armaanss: Destruction complete after 1s
google_cloud_identity_group_membership.tech_jazang: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/110468009800586964384]
google_cloud_identity_group_membership.tech_ethanli: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/105151532732144732168]

Apply complete! Resources: 2 added, 0 changed, 1 destroyed.

3. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
googleplay_user.ethanli: Creating...
googleplay_user.jazang: Creating...
googleplay_user.ethanli: Creation complete after 0s [id=ethanli@andrew.cmu.edu]
googleplay_user.jazang: Creation complete after 0s [id=jazang@andrew.cmu.edu]
googleplay_app_iam.terrier_ethanli: Creating...
googleplay_app_iam.terrier_jazang: Creating...
googleplay_app_iam.terrier_ethanli: Creation complete after 1s
googleplay_app_iam.terrier_jazang: Creation complete after 1s

Apply complete! Resources: 4 added, 0 changed, 0 destroyed.

4. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.23s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.armaanss: Destroying... [id=1353118209024331948:639282991192866826]
discord_member_roles.jazang: Creating...
discord_member_roles.ethanli: Creating...
discord_member_roles.armaanss: Destruction complete after 0s
discord_member_roles.jazang: Creation complete after 1s [id=1353118209024331948:1412952264556941466]
discord_member_roles.ethanli: Creation complete after 2s [id=1353118209024331948:770051855857221693]

Apply complete! Resources: 2 added, 0 changed, 1 destroyed.

5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.cmu-study_armaanss: Destroying...
forgejo_team_member.terrier_jazang: Creating...
forgejo_team_member.terrier_ethanli: Creating...
forgejo_team_member.terrier_jazang: Creation complete after 0s
forgejo_team_member.cmu-study_armaanss: Destruction complete after 0s
forgejo_team_member.terrier_ethanli: Creation complete after 0s
github_team_membership.terrier_jazang: Creating...
github_team_membership.terrier_ethanli: Creating...
github_team_membership.terrier_ethanli: Creation complete after 3s [id=12028577:elsaber505]
github_team_membership.terrier_jazang: Creation complete after 4s [id=12028577:jacqzang]

Apply complete! Resources: 4 added, 0 changed, 1 destroyed.

6. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.10s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cmu-study_study: Modifying... [id=scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a]
keycloak_group_memberships.project_terrier: Modifying... [id=scottylabs/group-memberships/678c7fa9-35cc-4bef-9a06-87cb6aa36238]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_cmu-study_study: Modifications complete after 0s [id=scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a]
keycloak_group_memberships.project_terrier: Modifications complete after 4s [id=scottylabs/group-memberships/678c7fa9-35cc-4bef-9a06-87cb6aa36238]
keycloak_group_memberships.projects: Modifications complete after 5s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 3 changed, 0 destroyed.

Apply Summary

6 projects, 6 successful, 0 failed, 0 errored

Ran Apply for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.23s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_hub_armaanss: Destroying... [id=7873933317574871801] null_resource.slack_cmu-study_ch0_armaanss: Destroying... [id=6301904129149314234] null_resource.slack_hub_armaanss: Destruction complete after 0s null_resource.slack_cmu-study_ch0_armaanss: Destruction complete after 0s null_resource.slack_terrier_ch0_ethanli: Creating... null_resource.slack_hub_ethanli: Creating... null_resource.slack_terrier_ch0_ethanli: Provisioning with 'local-exec'... null_resource.slack_terrier_ch0_ethanli (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08AAADPCSE --user U09F9AJBQAD"] null_resource.slack_hub_jazang: Creating... null_resource.slack_terrier_ch0_jazang: Creating... null_resource.slack_hub_ethanli: Provisioning with 'local-exec'... null_resource.slack_hub_ethanli (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U09F9AJBQAD"] null_resource.slack_terrier_ch0_jazang: Provisioning with 'local-exec'... null_resource.slack_terrier_ch0_jazang (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08AAADPCSE --user U0AAAKB971A"] null_resource.slack_hub_jazang: Provisioning with 'local-exec'... null_resource.slack_hub_jazang (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0AAAKB971A"] null_resource.slack_terrier_ch0_ethanli: Creation complete after 0s [id=2370713295190598071] null_resource.slack_hub_ethanli: Creation complete after 0s [id=8191639316496535190] null_resource.slack_hub_jazang: Creation complete after 0s [id=8147252572916099479] null_resource.slack_terrier_ch0_jazang: Creation complete after 0s [id=6186309242297098640] Apply complete! Resources: 4 added, 0 changed, 2 destroyed. ``` </details> --- ### 2. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.19s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_armaanss: Destroying... [id=groups/04anzqyu1bqbv79/memberships/115946035627561359960] google_cloud_identity_group_membership.tech_ethanli: Creating... google_cloud_identity_group_membership.tech_jazang: Creating... google_cloud_identity_group_membership.tech_armaanss: Destruction complete after 1s google_cloud_identity_group_membership.tech_jazang: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/110468009800586964384] google_cloud_identity_group_membership.tech_ethanli: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/105151532732144732168] Apply complete! Resources: 2 added, 0 changed, 1 destroyed. ``` </details> --- ### 3. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ googleplay_user.ethanli: Creating... googleplay_user.jazang: Creating... googleplay_user.ethanli: Creation complete after 0s [id=ethanli@andrew.cmu.edu] googleplay_user.jazang: Creation complete after 0s [id=jazang@andrew.cmu.edu] googleplay_app_iam.terrier_ethanli: Creating... googleplay_app_iam.terrier_jazang: Creating... googleplay_app_iam.terrier_ethanli: Creation complete after 1s googleplay_app_iam.terrier_jazang: Creation complete after 1s Apply complete! Resources: 4 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.23s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.armaanss: Destroying... [id=1353118209024331948:639282991192866826] discord_member_roles.jazang: Creating... discord_member_roles.ethanli: Creating... discord_member_roles.armaanss: Destruction complete after 0s discord_member_roles.jazang: Creation complete after 1s [id=1353118209024331948:1412952264556941466] discord_member_roles.ethanli: Creation complete after 2s [id=1353118209024331948:770051855857221693] Apply complete! Resources: 2 added, 0 changed, 1 destroyed. ``` </details> --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.cmu-study_armaanss: Destroying... forgejo_team_member.terrier_jazang: Creating... forgejo_team_member.terrier_ethanli: Creating... forgejo_team_member.terrier_jazang: Creation complete after 0s forgejo_team_member.cmu-study_armaanss: Destruction complete after 0s forgejo_team_member.terrier_ethanli: Creation complete after 0s github_team_membership.terrier_jazang: Creating... github_team_membership.terrier_ethanli: Creating... github_team_membership.terrier_ethanli: Creation complete after 3s [id=12028577:elsaber505] github_team_membership.terrier_jazang: Creation complete after 4s [id=12028577:jacqzang] Apply complete! Resources: 4 added, 0 changed, 1 destroyed. ``` </details> --- ### 6. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.10s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cmu-study_study: Modifying... [id=scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a] keycloak_group_memberships.project_terrier: Modifying... [id=scottylabs/group-memberships/678c7fa9-35cc-4bef-9a06-87cb6aa36238] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_cmu-study_study: Modifications complete after 0s [id=scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a] keycloak_group_memberships.project_terrier: Modifications complete after 4s [id=scottylabs/group-memberships/678c7fa9-35cc-4bef-9a06-87cb6aa36238] keycloak_group_memberships.projects: Modifications complete after 5s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 3 changed, 0 destroyed. ``` </details> --- ### Apply Summary 6 projects, 6 successful, 0 failed, 0 errored
anish merged commit f7add78e66 into main 2026-09-22 10:25:41 +00:00
anish deleted branch add-terrier-unapproved 2026-09-22 10:25:42 +00:00

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
Sign in to join this conversation.
No description provided.