feat: add armaanss and bxiang to cmu-study #339
No reviewers
Labels
No labels
needs/linking
needs/docs
needs/research
needs/upstream
type/bug
type/external
type/feature
type/refactor
urgency
high
urgency
immediate
urgency
low
urgency
medium
No milestone
No project
No assignees
3 participants
Notifications
Due date
No due date set.
Dependencies
No dependencies set
Reference
ScottyLabs/governance!339
Loading…
Reference in a new issue
No description provided.
Delete branch "add-cmu-study-r2"
Deleting a branch is permanent. Although the deleted branch may continue to exist for a short time before it actually gets removed, it CANNOT be undone in most cases. Continue?
Consolidates #269 #277
9dde0bf7715cadfb75ebatlantis plan
Ran Plan for 0 projects:
5cadfb75eb2ce4ca88a0Ran Plan for 6 projects:
google_playdir:tofu/google_playworkspace:defaultgitdir:tofu/gitworkspace:defaultslackdir:tofu/slackworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:default1. project:
google_playdir:tofu/google_playworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #308. To continue, delete the lock from #308 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #308. To continue, delete the lock from #308 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #308. To continue, delete the lock from #308 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #308. To continue, delete the lock from #308 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #308. To continue, delete the lock from #308 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.6. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #308. To continue, delete the lock from #308 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.Plan Summary
6 projects, 0 with changes, 0 with no changes, 6 failed
atlantis unlock
All Atlantis locks for this PR have been unlocked and plans discarded
atlantis plan
Ran Plan for 6 projects:
slackdir:tofu/slackworkspace:defaultgitdir:tofu/gitworkspace:defaultgoogle_playdir:tofu/google_playworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultdiscorddir:tofu/discordworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #308. To continue, delete the lock from #308 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #308. To continue, delete the lock from #308 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
google_playdir:tofu/google_playworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #308. To continue, delete the lock from #308 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #308. To continue, delete the lock from #308 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #308. To continue, delete the lock from #308 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.6. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #308. To continue, delete the lock from #308 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.Plan Summary
6 projects, 0 with changes, 0 with no changes, 6 failed
Ran Plan for 6 projects:
slackdir:tofu/slackworkspace:defaultgoogle_playdir:tofu/google_playworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 15 to destroy.
2. project:
google_playdir:tofu/google_playworkspace:defaultShow Output
Plan: 0 to add, 0 to change, 14 to destroy.
3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 8 to destroy.
4. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 0 to add, 4 to change, 0 to destroy.
5. project:
discorddir:tofu/discordworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 6 to destroy.
6. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 16 to destroy.
Plan Summary
6 projects, 6 with changes, 0 with no changes, 0 failed
atlantis apply
Ran Apply for 6 projects:
slackdir:tofu/slackworkspace:defaultgoogle_playdir:tofu/google_playworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
2. project:
google_playdir:tofu/google_playworkspace:defaultShow Output
3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
4. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
5. project:
discorddir:tofu/discordworkspace:defaultShow Output
6. project:
gitdir:tofu/gitworkspace:defaultApply Error
Show Output
Apply Summary
6 projects, 5 successful, 0 failed, 1 errored
@armaanss you need 2FA on github
atlantis unlock
All Atlantis locks for this PR have been unlocked and plans discarded
ed6cb0f8ce91efca93e5Ran Plan for 5 projects:
slackdir:tofu/slackworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 0 to destroy.
2. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 0 to add, 2 to change, 0 to destroy.
3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 0 to destroy.
4. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 0 to destroy.
5. project:
discorddir:tofu/discordworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 0 to destroy.
Plan Summary
5 projects, 5 with changes, 0 with no changes, 0 failed
91efca93e5d036b46134feat: add armaanss, bxiang, ocervera to cmu-studyto feat: add armaanss and bxiang to cmu-studyPlan Error
Ran Plan for 5 projects:
slackdir:tofu/slackworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultdiscorddir:tofu/discordworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultPlan Error
Show Output
2. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Error
Show Output
3. project:
discorddir:tofu/discordworkspace:defaultPlan Error
Show Output
4. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Error
Show Output
5. project:
gitdir:tofu/gitworkspace:defaultPlan Error
Show Output
Plan Summary
5 projects, 0 with changes, 0 with no changes, 5 failed
d036b4613402730b6904Plan Error
Ran Plan for 5 projects:
gitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:defaultslackdir:tofu/slackworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultkeycloakdir:tofu/keycloakworkspace:default1. project:
gitdir:tofu/gitworkspace:defaultPlan Error
Show Output
2. project:
discorddir:tofu/discordworkspace:defaultPlan Error
Show Output
3. project:
slackdir:tofu/slackworkspace:defaultPlan Error
Show Output
4. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Error
Show Output
5. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Error
Show Output
Plan Summary
5 projects, 0 with changes, 0 with no changes, 5 failed
02730b69046f0b5eca16Ran Plan for 5 projects:
slackdir:tofu/slackworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultPlan Error
Show Output
2. project:
discorddir:tofu/discordworkspace:defaultPlan Error
Show Output
3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Error
Show Output
4. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Error
Show Output
5. project:
gitdir:tofu/gitworkspace:defaultPlan Error
Show Output
Plan Summary
5 projects, 0 with changes, 0 with no changes, 5 failed
@bxiang see step 1
atlantis unlock
All Atlantis locks for this PR have been unlocked and plans discarded
Consolidated into #356. Closing.
Pull request closed