feat: restore gostmeaper #329

Merged
xboxbedrock merged 2 commits from gostmeaper/governance:main into main 2026-09-21 03:01:05 +00:00
Member
No description provided.
feat: ADD ME BACK
All checks were successful
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 4m20s
CI / check-1 (pull_request) Successful in 5m51s
CI / check (pull_request) Successful in 0s
aba5990ad2
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 4 to change, 0 to destroy.
atlantis/plan: discord Plan: 0 to add, 1 to change, 0 to destroy.
atlantis/plan: git Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan 4/4 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 4/4 projects applied successfully.
bfeb68aa74

Ran Plan for 4 projects:

  1. project: git dir: tofu/git workspace: default
  2. project: keycloak dir: tofu/keycloak workspace: default
  3. project: slack dir: tofu/slack workspace: default
  4. project: discord dir: tofu/discord workspace: default

1. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

4 projects, 0 with changes, 0 with no changes, 4 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 4 projects: 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` --- ### 1. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 4 projects, 0 with changes, 0 with no changes, 4 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
tentype left a comment

Not today nuh uh

Not today nuh uh
Author
Member

I'm taking this up with HR

I'm taking this up with HR
Member

atlantis plan

atlantis plan
xboxbedrock changed title from feat: ADD ME BACK to feat: restore gostmeaper 2026-09-21 02:58:20 +00:00

Plan Error

cannot run "plan": pull request 329 is currently locked by "plan" for commit bfeb68a.
Wait until the previous command is complete and try again
**Plan Error** ``` cannot run "plan": pull request 329 is currently locked by "plan" for commit bfeb68a. Wait until the previous command is complete and try again ``` - Commit: [bfeb68aa749e993714fdc112649ccbd48d3b7608](https://git.cmu.dev/ScottyLabs/governance/pulls/329/commits/bfeb68aa749e993714fdc112649ccbd48d3b7608) - Multiple Atlantis jobs are currently running for this commit.

Ran Plan for 4 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: keycloak dir: tofu/keycloak workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on shared package cache
   Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/329/default/crates/governance-schema)
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/329/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/329/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/329/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 8.04s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_cbp_ch0_gostmeaper will be created
+ resource "null_resource" "slack_cbp_ch0_gostmeaper" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08S56CJ8R2"
          + "user"    = "U09DV9S4RDZ"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 1 to add, 0 to change, 0 to destroy.


2. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 8.05s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cbp_bus-sign will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_bus-sign" {
        id       = "scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b"
!       members  = [
+           "johncao",
            # (13 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_dalmatian will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_dalmatian" {
        id       = "scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153"
!       members  = [
+           "johncao",
            # (13 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_discord-verify will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_discord-verify" {
        id       = "scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745"
!       members  = [
+           "johncao",
            # (13 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_groupme-mirror will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_groupme-mirror" {
        id       = "scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39"
!       members  = [
+           "johncao",
            # (13 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 4 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 4 to change, 0 to destroy.


3. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 8.08s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # discord_member_roles.gostmeaper will be updated in-place
! resource "discord_member_roles" "gostmeaper" {
        id        = "1353118209024331948:459877032176779264"
        # (2 unchanged attributes hidden)

+       role {
+           has_role = true
+           role_id  = "1494773174334783499"
        }

        # (3 unchanged blocks hidden)
    }

Plan: 0 to add, 1 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 0 to add, 1 to change, 0 to destroy.


4. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 8.06s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # forgejo_team_member.cbp_gostmeaper will be created
+ resource "forgejo_team_member" "cbp_gostmeaper" {
+       team_id = 3
+       user    = "johncao"
    }

  # github_team_membership.cbp_gostmeaper will be created
+ resource "github_team_membership" "cbp_gostmeaper" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "17170099"
+       username = "gostmeaper"
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 0 to change, 0 to destroy.


Plan Summary

4 projects, 4 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 4 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on shared package cache Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/329/default/crates/governance-schema) Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/329/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/329/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/329/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 8.04s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_cbp_ch0_gostmeaper will be created + resource "null_resource" "slack_cbp_ch0_gostmeaper" { + id = (known after apply) + triggers = { + "channel" = "C08S56CJ8R2" + "user" = "U09DV9S4RDZ" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 2. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 8.05s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cbp_bus-sign will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_bus-sign" { id = "scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b" ! members = [ + "johncao", # (13 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_dalmatian will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_dalmatian" { id = "scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153" ! members = [ + "johncao", # (13 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_discord-verify will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_discord-verify" { id = "scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745" ! members = [ + "johncao", # (13 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_groupme-mirror will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_groupme-mirror" { id = "scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39" ! members = [ + "johncao", # (13 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 4 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 4 to change, 0 to destroy. --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 8.08s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # discord_member_roles.gostmeaper will be updated in-place ! resource "discord_member_roles" "gostmeaper" { id = "1353118209024331948:459877032176779264" # (2 unchanged attributes hidden) + role { + has_role = true + role_id = "1494773174334783499" } # (3 unchanged blocks hidden) } Plan: 0 to add, 1 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 0 to add, 1 to change, 0 to destroy. --- ### 4. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 8.06s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # forgejo_team_member.cbp_gostmeaper will be created + resource "forgejo_team_member" "cbp_gostmeaper" { + team_id = 3 + user = "johncao" } # github_team_membership.cbp_gostmeaper will be created + resource "github_team_membership" "cbp_gostmeaper" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "17170099" + username = "gostmeaper" } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### Plan Summary 4 projects, 4 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis apply

atlantis apply

Ran Apply for 4 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: git dir: tofu/git workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.11s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_cbp_ch0_gostmeaper: Creating...
null_resource.slack_cbp_ch0_gostmeaper: Provisioning with 'local-exec'...
null_resource.slack_cbp_ch0_gostmeaper (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08S56CJ8R2 --user U09DV9S4RDZ"]
null_resource.slack_cbp_ch0_gostmeaper: Creation complete after 0s [id=922029860556332272]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

2. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.12s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.gostmeaper: Modifying... [id=1353118209024331948:459877032176779264]
discord_member_roles.gostmeaper: Modifications complete after 0s [id=1353118209024331948:459877032176779264]

Apply complete! Resources: 0 added, 1 changed, 0 destroyed.

3. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.13s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.cbp_gostmeaper: Creating...
forgejo_team_member.cbp_gostmeaper: Creation complete after 0s
github_team_membership.cbp_gostmeaper: Creating...
github_team_membership.cbp_gostmeaper: Creation complete after 1s [id=17170099:gostmeaper]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.14s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cbp_groupme-mirror: Modifying... [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39]
keycloak_group_memberships.project_cbp_dalmatian: Modifying... [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153]
keycloak_group_memberships.project_cbp_discord-verify: Modifying... [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745]
keycloak_group_memberships.project_cbp_bus-sign: Modifying... [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b]
keycloak_group_memberships.project_cbp_dalmatian: Modifications complete after 2s [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153]
keycloak_group_memberships.project_cbp_discord-verify: Modifications complete after 2s [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745]
keycloak_group_memberships.project_cbp_groupme-mirror: Modifications complete after 2s [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39]
keycloak_group_memberships.project_cbp_bus-sign: Modifications complete after 2s [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b]

Apply complete! Resources: 0 added, 4 changed, 0 destroyed.

Apply Summary

4 projects, 4 successful, 0 failed, 0 errored

Ran Apply for 4 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.11s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_cbp_ch0_gostmeaper: Creating... null_resource.slack_cbp_ch0_gostmeaper: Provisioning with 'local-exec'... null_resource.slack_cbp_ch0_gostmeaper (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08S56CJ8R2 --user U09DV9S4RDZ"] null_resource.slack_cbp_ch0_gostmeaper: Creation complete after 0s [id=922029860556332272] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.12s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.gostmeaper: Modifying... [id=1353118209024331948:459877032176779264] discord_member_roles.gostmeaper: Modifications complete after 0s [id=1353118209024331948:459877032176779264] Apply complete! Resources: 0 added, 1 changed, 0 destroyed. ``` </details> --- ### 3. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.13s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.cbp_gostmeaper: Creating... forgejo_team_member.cbp_gostmeaper: Creation complete after 0s github_team_membership.cbp_gostmeaper: Creating... github_team_membership.cbp_gostmeaper: Creation complete after 1s [id=17170099:gostmeaper] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.14s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cbp_groupme-mirror: Modifying... [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39] keycloak_group_memberships.project_cbp_dalmatian: Modifying... [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153] keycloak_group_memberships.project_cbp_discord-verify: Modifying... [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745] keycloak_group_memberships.project_cbp_bus-sign: Modifying... [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b] keycloak_group_memberships.project_cbp_dalmatian: Modifications complete after 2s [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153] keycloak_group_memberships.project_cbp_discord-verify: Modifications complete after 2s [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745] keycloak_group_memberships.project_cbp_groupme-mirror: Modifications complete after 2s [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39] keycloak_group_memberships.project_cbp_bus-sign: Modifications complete after 2s [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b] Apply complete! Resources: 0 added, 4 changed, 0 destroyed. ``` </details> --- ### Apply Summary 4 projects, 4 successful, 0 failed, 0 errored

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
xboxbedrock referenced this pull request from a commit 2026-09-21 03:01:06 +00:00
brightz referenced this pull request from a commit 2026-09-26 22:03:28 +00:00
Sign in to join this conversation.
No description provided.