feat: jiya and jesse as leads for PR approvals #298

Merged
thesuperRL merged 2 commits from krishsax/governance:main into main 2026-09-20 00:28:25 +00:00
Member
No description provided.
feat: jiya and jesse as leads for PR approvals
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
Validate / Governance (pull_request_target) Successful in 8m50s
CI / build (pull_request) Has been skipped
CI / check-1 (pull_request) Successful in 21m37s
CI / check (pull_request) Successful in 0s
eb7090fa98
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 4 to change, 0 to destroy.
atlantis/plan: discord Plan: 0 to add, 2 to change, 0 to destroy.
atlantis/plan: git Plan: 10 to add, 2 to change, 0 to destroy.
atlantis/plan 5/5 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 5/5 projects applied successfully.
a6769d606d

Ran Plan for 5 projects:

  1. project: git dir: tofu/git workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: slack dir: tofu/slack workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: google_groups dir: tofu/google_groups workspace: default

1. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #274. To continue, delete the lock from #274 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #274. To continue, delete the lock from #274 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #274. To continue, delete the lock from #274 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #274. To continue, delete the lock from #274 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #274. To continue, delete the lock from #274 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

5 projects, 0 with changes, 0 with no changes, 5 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 5 projects: 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` --- ### 1. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #274. To continue, delete the lock from #274 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #274. To continue, delete the lock from #274 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #274. To continue, delete the lock from #274 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #274. To continue, delete the lock from #274 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #274. To continue, delete the lock from #274 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 5 projects, 0 with changes, 0 with no changes, 5 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis plan

atlantis plan

Ran Plan for 5 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_groups dir: tofu/google_groups workspace: default
  3. project: keycloak dir: tofu/keycloak workspace: default
  4. project: discord dir: tofu/discord workspace: default
  5. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
   Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/298/default/crates/governance-schema)
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/298/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/298/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/298/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 9.05s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_leads_jc_965 will be created
+ resource "null_resource" "slack_leads_jc_965" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0AN1M0HDV0"
          + "user"    = "U09E19AABEY"
        }
    }

  # null_resource.slack_leads_jiyas will be created
+ resource "null_resource" "slack_leads_jiyas" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0AN1M0HDV0"
          + "user"    = "U09F6FZH78A"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 0 to destroy.


2. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.06s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.ops_jc_965 will be created
+ resource "google_cloud_identity_group_membership" "ops_jc_965" {
+       create_time = (known after apply)
+       group       = "groups/03fwokq036zsp1f"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "jessec2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

  # google_cloud_identity_group_membership.ops_jiyas will be created
+ resource "google_cloud_identity_group_membership" "ops_jiyas" {
+       create_time = (known after apply)
+       group       = "groups/03fwokq036zsp1f"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "jiyas@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 2 to add, 0 to change, 0 to destroy.


3. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.06s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_slai_cmugpt-agent_admins will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-agent_admins" {
        id       = "scottylabs/group-memberships/f4e82f9b-769d-4f38-a7f2-e214189d9269"
!       members  = [
+           "jessec2",
+           "jiyas",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-mcp-server_admins will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server_admins" {
        id       = "scottylabs/group-memberships/8f84f9ec-e78a-41f6-9ce6-097fce273298"
!       members  = [
+           "jessec2",
+           "jiyas",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-sms-surface_admins will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface_admins" {
        id       = "scottylabs/group-memberships/34404457-985f-4a61-b2f8-4754c2eb3293"
!       members  = [
+           "jessec2",
+           "jiyas",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-surface_admins will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-surface_admins" {
        id       = "scottylabs/group-memberships/4f3531db-8075-480e-bbf4-253e98783b56"
!       members  = [
+           "jessec2",
+           "jiyas",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 4 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 4 to change, 0 to destroy.


4. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.06s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # discord_member_roles.jc_965 will be updated in-place
! resource "discord_member_roles" "jc_965" {
        id        = "1353118209024331948:472617578938761226"
        # (2 unchanged attributes hidden)

+       role {
+           has_role = true
+           role_id  = "1519792320755662928"
        }

        # (2 unchanged blocks hidden)
    }

  # discord_member_roles.jiyas will be updated in-place
! resource "discord_member_roles" "jiyas" {
        id        = "1353118209024331948:829775805768925275"
        # (2 unchanged attributes hidden)

+       role {
+           has_role = true
+           role_id  = "1519792320755662928"
        }

        # (2 unchanged blocks hidden)
    }

Plan: 0 to add, 2 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 0 to add, 2 to change, 0 to destroy.


5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.09s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # forgejo_team_member.cmugpt-agent_leads_jc_965 will be created
+ resource "forgejo_team_member" "cmugpt-agent_leads_jc_965" {
+       team_id = 24
+       user    = "jessec2"
    }

  # forgejo_team_member.cmugpt-agent_leads_jiyas will be created
+ resource "forgejo_team_member" "cmugpt-agent_leads_jiyas" {
+       team_id = 24
+       user    = "jiyas"
    }

  # forgejo_team_member.cmugpt-mcp-server_leads_jc_965 will be created
+ resource "forgejo_team_member" "cmugpt-mcp-server_leads_jc_965" {
+       team_id = 26
+       user    = "jessec2"
    }

  # forgejo_team_member.cmugpt-mcp-server_leads_jiyas will be created
+ resource "forgejo_team_member" "cmugpt-mcp-server_leads_jiyas" {
+       team_id = 26
+       user    = "jiyas"
    }

  # forgejo_team_member.cmugpt-sms-surface_leads_jc_965 will be created
+ resource "forgejo_team_member" "cmugpt-sms-surface_leads_jc_965" {
+       team_id = 25
+       user    = "jessec2"
    }

  # forgejo_team_member.cmugpt-sms-surface_leads_jiyas will be created
+ resource "forgejo_team_member" "cmugpt-sms-surface_leads_jiyas" {
+       team_id = 25
+       user    = "jiyas"
    }

  # forgejo_team_member.cmugpt-surface_leads_jc_965 will be created
+ resource "forgejo_team_member" "cmugpt-surface_leads_jc_965" {
+       team_id = 23
+       user    = "jessec2"
    }

  # forgejo_team_member.cmugpt-surface_leads_jiyas will be created
+ resource "forgejo_team_member" "cmugpt-surface_leads_jiyas" {
+       team_id = 23
+       user    = "jiyas"
    }

  # forgejo_team_member.tech_leads_jc_965 will be created
+ resource "forgejo_team_member" "tech_leads_jc_965" {
+       team_id = 44
+       user    = "jessec2"
    }

  # forgejo_team_member.tech_leads_jiyas will be created
+ resource "forgejo_team_member" "tech_leads_jiyas" {
+       team_id = 44
+       user    = "jiyas"
    }

  # github_team_membership.slai_jc_965 will be updated in-place
! resource "github_team_membership" "slai_jc_965" {
        id       = "17571328:jc-965"
!       role     = "member" -> "maintainer"
        # (3 unchanged attributes hidden)
    }

  # github_team_membership.slai_jiyas will be updated in-place
! resource "github_team_membership" "slai_jiyas" {
        id       = "17571328:jiya2301"
!       role     = "member" -> "maintainer"
        # (3 unchanged attributes hidden)
    }

Plan: 10 to add, 2 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 10 to add, 2 to change, 0 to destroy.


Plan Summary

5 projects, 5 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 5 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/298/default/crates/governance-schema) Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/298/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/298/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/298/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 9.05s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_leads_jc_965 will be created + resource "null_resource" "slack_leads_jc_965" { + id = (known after apply) + triggers = { + "channel" = "C0AN1M0HDV0" + "user" = "U09E19AABEY" } } # null_resource.slack_leads_jiyas will be created + resource "null_resource" "slack_leads_jiyas" { + id = (known after apply) + triggers = { + "channel" = "C0AN1M0HDV0" + "user" = "U09F6FZH78A" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 2. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.06s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.ops_jc_965 will be created + resource "google_cloud_identity_group_membership" "ops_jc_965" { + create_time = (known after apply) + group = "groups/03fwokq036zsp1f" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "jessec2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } # google_cloud_identity_group_membership.ops_jiyas will be created + resource "google_cloud_identity_group_membership" "ops_jiyas" { + create_time = (known after apply) + group = "groups/03fwokq036zsp1f" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "jiyas@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 3. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.06s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_slai_cmugpt-agent_admins will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-agent_admins" { id = "scottylabs/group-memberships/f4e82f9b-769d-4f38-a7f2-e214189d9269" ! members = [ + "jessec2", + "jiyas", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-mcp-server_admins will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server_admins" { id = "scottylabs/group-memberships/8f84f9ec-e78a-41f6-9ce6-097fce273298" ! members = [ + "jessec2", + "jiyas", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-sms-surface_admins will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface_admins" { id = "scottylabs/group-memberships/34404457-985f-4a61-b2f8-4754c2eb3293" ! members = [ + "jessec2", + "jiyas", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-surface_admins will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-surface_admins" { id = "scottylabs/group-memberships/4f3531db-8075-480e-bbf4-253e98783b56" ! members = [ + "jessec2", + "jiyas", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 4 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 4 to change, 0 to destroy. --- ### 4. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.06s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # discord_member_roles.jc_965 will be updated in-place ! resource "discord_member_roles" "jc_965" { id = "1353118209024331948:472617578938761226" # (2 unchanged attributes hidden) + role { + has_role = true + role_id = "1519792320755662928" } # (2 unchanged blocks hidden) } # discord_member_roles.jiyas will be updated in-place ! resource "discord_member_roles" "jiyas" { id = "1353118209024331948:829775805768925275" # (2 unchanged attributes hidden) + role { + has_role = true + role_id = "1519792320755662928" } # (2 unchanged blocks hidden) } Plan: 0 to add, 2 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 0 to add, 2 to change, 0 to destroy. --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.09s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # forgejo_team_member.cmugpt-agent_leads_jc_965 will be created + resource "forgejo_team_member" "cmugpt-agent_leads_jc_965" { + team_id = 24 + user = "jessec2" } # forgejo_team_member.cmugpt-agent_leads_jiyas will be created + resource "forgejo_team_member" "cmugpt-agent_leads_jiyas" { + team_id = 24 + user = "jiyas" } # forgejo_team_member.cmugpt-mcp-server_leads_jc_965 will be created + resource "forgejo_team_member" "cmugpt-mcp-server_leads_jc_965" { + team_id = 26 + user = "jessec2" } # forgejo_team_member.cmugpt-mcp-server_leads_jiyas will be created + resource "forgejo_team_member" "cmugpt-mcp-server_leads_jiyas" { + team_id = 26 + user = "jiyas" } # forgejo_team_member.cmugpt-sms-surface_leads_jc_965 will be created + resource "forgejo_team_member" "cmugpt-sms-surface_leads_jc_965" { + team_id = 25 + user = "jessec2" } # forgejo_team_member.cmugpt-sms-surface_leads_jiyas will be created + resource "forgejo_team_member" "cmugpt-sms-surface_leads_jiyas" { + team_id = 25 + user = "jiyas" } # forgejo_team_member.cmugpt-surface_leads_jc_965 will be created + resource "forgejo_team_member" "cmugpt-surface_leads_jc_965" { + team_id = 23 + user = "jessec2" } # forgejo_team_member.cmugpt-surface_leads_jiyas will be created + resource "forgejo_team_member" "cmugpt-surface_leads_jiyas" { + team_id = 23 + user = "jiyas" } # forgejo_team_member.tech_leads_jc_965 will be created + resource "forgejo_team_member" "tech_leads_jc_965" { + team_id = 44 + user = "jessec2" } # forgejo_team_member.tech_leads_jiyas will be created + resource "forgejo_team_member" "tech_leads_jiyas" { + team_id = 44 + user = "jiyas" } # github_team_membership.slai_jc_965 will be updated in-place ! resource "github_team_membership" "slai_jc_965" { id = "17571328:jc-965" ! role = "member" -> "maintainer" # (3 unchanged attributes hidden) } # github_team_membership.slai_jiyas will be updated in-place ! resource "github_team_membership" "slai_jiyas" { id = "17571328:jiya2301" ! role = "member" -> "maintainer" # (3 unchanged attributes hidden) } Plan: 10 to add, 2 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 10 to add, 2 to change, 0 to destroy. --- ### Plan Summary 5 projects, 5 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis apply

atlantis apply

Ran Apply for 5 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: git dir: tofu/git workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.15s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_leads_jc_965: Creating...
null_resource.slack_leads_jc_965: Provisioning with 'local-exec'...
null_resource.slack_leads_jc_965 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0AN1M0HDV0 --user U09E19AABEY"]
null_resource.slack_leads_jiyas: Creating...
null_resource.slack_leads_jiyas: Provisioning with 'local-exec'...
null_resource.slack_leads_jiyas (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0AN1M0HDV0 --user U09F6FZH78A"]
null_resource.slack_leads_jiyas: Creation complete after 0s [id=1866307958230487938]
null_resource.slack_leads_jc_965: Creation complete after 0s [id=6185492700614965496]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

2. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.12s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.jc_965: Modifying... [id=1353118209024331948:472617578938761226]
discord_member_roles.jiyas: Modifying... [id=1353118209024331948:829775805768925275]
discord_member_roles.jiyas: Modifications complete after 0s [id=1353118209024331948:829775805768925275]
discord_member_roles.jc_965: Modifications complete after 1s [id=1353118209024331948:472617578938761226]

Apply complete! Resources: 0 added, 2 changed, 0 destroyed.

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.17s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.ops_jiyas: Creating...
google_cloud_identity_group_membership.ops_jc_965: Creating...
google_cloud_identity_group_membership.ops_jiyas: Creation complete after 1s [id=groups/03fwokq036zsp1f/memberships/116261093135461470038]
google_cloud_identity_group_membership.ops_jc_965: Creation complete after 1s [id=groups/03fwokq036zsp1f/memberships/102552619255005499474]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

4. project: git dir: tofu/git workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.05s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.tech_leads_jiyas: Creating...
forgejo_team_member.cmugpt-agent_leads_jc_965: Creating...
forgejo_team_member.cmugpt-mcp-server_leads_jc_965: Creating...
forgejo_team_member.cmugpt-sms-surface_leads_jc_965: Creating...
forgejo_team_member.cmugpt-agent_leads_jiyas: Creating...
forgejo_team_member.cmugpt-surface_leads_jiyas: Creating...
forgejo_team_member.cmugpt-surface_leads_jc_965: Creating...
forgejo_team_member.tech_leads_jc_965: Creating...
forgejo_team_member.cmugpt-mcp-server_leads_jiyas: Creating...
forgejo_team_member.cmugpt-agent_leads_jiyas: Creation complete after 0s
forgejo_team_member.tech_leads_jiyas: Creation complete after 0s
forgejo_team_member.cmugpt-sms-surface_leads_jiyas: Creating...
forgejo_team_member.cmugpt-mcp-server_leads_jc_965: Creation complete after 0s
forgejo_team_member.cmugpt-surface_leads_jiyas: Creation complete after 0s
forgejo_team_member.cmugpt-mcp-server_leads_jiyas: Creation complete after 0s
forgejo_team_member.cmugpt-sms-surface_leads_jc_965: Creation complete after 0s
forgejo_team_member.cmugpt-agent_leads_jc_965: Creation complete after 0s
forgejo_team_member.cmugpt-surface_leads_jc_965: Creation complete after 0s
forgejo_team_member.tech_leads_jc_965: Creation complete after 0s
forgejo_team_member.cmugpt-sms-surface_leads_jiyas: Creation complete after 0s
github_team_membership.slai_jc_965: Modifying... [id=17571328:jc-965]
github_team_membership.slai_jiyas: Modifying... [id=17571328:jiya2301]
github_team_membership.slai_jc_965: Modifications complete after 3s [id=17571328:jc-965]
github_team_membership.slai_jiyas: Modifications complete after 3s [id=17571328:jiya2301]

Apply complete! Resources: 10 added, 2 changed, 0 destroyed.

5. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.15s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_slai_cmugpt-mcp-server_admins: Modifying... [id=scottylabs/group-memberships/8f84f9ec-e78a-41f6-9ce6-097fce273298]
keycloak_group_memberships.project_slai_cmugpt-sms-surface_admins: Modifying... [id=scottylabs/group-memberships/34404457-985f-4a61-b2f8-4754c2eb3293]
keycloak_group_memberships.project_slai_cmugpt-surface_admins: Modifying... [id=scottylabs/group-memberships/4f3531db-8075-480e-bbf4-253e98783b56]
keycloak_group_memberships.project_slai_cmugpt-agent_admins: Modifying... [id=scottylabs/group-memberships/f4e82f9b-769d-4f38-a7f2-e214189d9269]
keycloak_group_memberships.project_slai_cmugpt-mcp-server_admins: Modifications complete after 4s [id=scottylabs/group-memberships/8f84f9ec-e78a-41f6-9ce6-097fce273298]
keycloak_group_memberships.project_slai_cmugpt-sms-surface_admins: Modifications complete after 4s [id=scottylabs/group-memberships/34404457-985f-4a61-b2f8-4754c2eb3293]
keycloak_group_memberships.project_slai_cmugpt-surface_admins: Modifications complete after 4s [id=scottylabs/group-memberships/4f3531db-8075-480e-bbf4-253e98783b56]
keycloak_group_memberships.project_slai_cmugpt-agent_admins: Modifications complete after 4s [id=scottylabs/group-memberships/f4e82f9b-769d-4f38-a7f2-e214189d9269]

Apply complete! Resources: 0 added, 4 changed, 0 destroyed.

Apply Summary

5 projects, 5 successful, 0 failed, 0 errored

Ran Apply for 5 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.15s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_leads_jc_965: Creating... null_resource.slack_leads_jc_965: Provisioning with 'local-exec'... null_resource.slack_leads_jc_965 (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0AN1M0HDV0 --user U09E19AABEY"] null_resource.slack_leads_jiyas: Creating... null_resource.slack_leads_jiyas: Provisioning with 'local-exec'... null_resource.slack_leads_jiyas (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0AN1M0HDV0 --user U09F6FZH78A"] null_resource.slack_leads_jiyas: Creation complete after 0s [id=1866307958230487938] null_resource.slack_leads_jc_965: Creation complete after 0s [id=6185492700614965496] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.12s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.jc_965: Modifying... [id=1353118209024331948:472617578938761226] discord_member_roles.jiyas: Modifying... [id=1353118209024331948:829775805768925275] discord_member_roles.jiyas: Modifications complete after 0s [id=1353118209024331948:829775805768925275] discord_member_roles.jc_965: Modifications complete after 1s [id=1353118209024331948:472617578938761226] Apply complete! Resources: 0 added, 2 changed, 0 destroyed. ``` </details> --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.17s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.ops_jiyas: Creating... google_cloud_identity_group_membership.ops_jc_965: Creating... google_cloud_identity_group_membership.ops_jiyas: Creation complete after 1s [id=groups/03fwokq036zsp1f/memberships/116261093135461470038] google_cloud_identity_group_membership.ops_jc_965: Creation complete after 1s [id=groups/03fwokq036zsp1f/memberships/102552619255005499474] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.05s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.tech_leads_jiyas: Creating... forgejo_team_member.cmugpt-agent_leads_jc_965: Creating... forgejo_team_member.cmugpt-mcp-server_leads_jc_965: Creating... forgejo_team_member.cmugpt-sms-surface_leads_jc_965: Creating... forgejo_team_member.cmugpt-agent_leads_jiyas: Creating... forgejo_team_member.cmugpt-surface_leads_jiyas: Creating... forgejo_team_member.cmugpt-surface_leads_jc_965: Creating... forgejo_team_member.tech_leads_jc_965: Creating... forgejo_team_member.cmugpt-mcp-server_leads_jiyas: Creating... forgejo_team_member.cmugpt-agent_leads_jiyas: Creation complete after 0s forgejo_team_member.tech_leads_jiyas: Creation complete after 0s forgejo_team_member.cmugpt-sms-surface_leads_jiyas: Creating... forgejo_team_member.cmugpt-mcp-server_leads_jc_965: Creation complete after 0s forgejo_team_member.cmugpt-surface_leads_jiyas: Creation complete after 0s forgejo_team_member.cmugpt-mcp-server_leads_jiyas: Creation complete after 0s forgejo_team_member.cmugpt-sms-surface_leads_jc_965: Creation complete after 0s forgejo_team_member.cmugpt-agent_leads_jc_965: Creation complete after 0s forgejo_team_member.cmugpt-surface_leads_jc_965: Creation complete after 0s forgejo_team_member.tech_leads_jc_965: Creation complete after 0s forgejo_team_member.cmugpt-sms-surface_leads_jiyas: Creation complete after 0s github_team_membership.slai_jc_965: Modifying... [id=17571328:jc-965] github_team_membership.slai_jiyas: Modifying... [id=17571328:jiya2301] github_team_membership.slai_jc_965: Modifications complete after 3s [id=17571328:jc-965] github_team_membership.slai_jiyas: Modifications complete after 3s [id=17571328:jiya2301] Apply complete! Resources: 10 added, 2 changed, 0 destroyed. ``` </details> --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.15s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_slai_cmugpt-mcp-server_admins: Modifying... [id=scottylabs/group-memberships/8f84f9ec-e78a-41f6-9ce6-097fce273298] keycloak_group_memberships.project_slai_cmugpt-sms-surface_admins: Modifying... [id=scottylabs/group-memberships/34404457-985f-4a61-b2f8-4754c2eb3293] keycloak_group_memberships.project_slai_cmugpt-surface_admins: Modifying... [id=scottylabs/group-memberships/4f3531db-8075-480e-bbf4-253e98783b56] keycloak_group_memberships.project_slai_cmugpt-agent_admins: Modifying... [id=scottylabs/group-memberships/f4e82f9b-769d-4f38-a7f2-e214189d9269] keycloak_group_memberships.project_slai_cmugpt-mcp-server_admins: Modifications complete after 4s [id=scottylabs/group-memberships/8f84f9ec-e78a-41f6-9ce6-097fce273298] keycloak_group_memberships.project_slai_cmugpt-sms-surface_admins: Modifications complete after 4s [id=scottylabs/group-memberships/34404457-985f-4a61-b2f8-4754c2eb3293] keycloak_group_memberships.project_slai_cmugpt-surface_admins: Modifications complete after 4s [id=scottylabs/group-memberships/4f3531db-8075-480e-bbf4-253e98783b56] keycloak_group_memberships.project_slai_cmugpt-agent_admins: Modifications complete after 4s [id=scottylabs/group-memberships/f4e82f9b-769d-4f38-a7f2-e214189d9269] Apply complete! Resources: 0 added, 4 changed, 0 destroyed. ``` </details> --- ### Apply Summary 5 projects, 5 successful, 0 failed, 0 errored

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
Sign in to join this conversation.
No description provided.