feat: add katelynk to cbp #288

Closed
katelynk wants to merge 2 commits from katelynk/governance:katelynk-patch-1 into main
Member
No description provided.
Update data/teams/cbp.toml
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Has been cancelled
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
CI / build (pull_request) Has been cancelled
0c06fa9ad1
xboxbedrock force-pushed katelynk-patch-1 from 0c06fa9ad1
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Has been cancelled
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
CI / build (pull_request) Has been cancelled
to a864b2c17d
Some checks failed
CI / check-1 (pull_request) Successful in 12m9s
CI / check (pull_request) Successful in 0s
Validate / Governance (pull_request_target) Failing after 8m27s
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
2026-09-19 21:38:08 +00:00
Compare
tentype requested changes 2026-09-20 00:42:02 +00:00
Dismissed
tentype left a comment

Your cmu.dev account is not linked, please link it here!

Once that's done, let me know and I'll approve. Thanks!

Your cmu.dev account is not linked, please link it [here](https://idp.scottylabs.org/realms/scottylabs/account/account-security/linked-accounts)! Once that's done, let me know and I'll approve. Thanks!
tentype force-pushed katelynk-patch-1 from a864b2c17d
Some checks failed
CI / check-1 (pull_request) Successful in 12m9s
CI / check (pull_request) Successful in 0s
Validate / Governance (pull_request_target) Failing after 8m27s
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
to ff8584e073
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan 0/0 projects planned successfully.
atlantis/apply 0/0 projects applied successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
Validate / Governance (pull_request_target) Successful in 4m54s
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
CI / build (pull_request) Has been cancelled
2026-09-20 21:21:13 +00:00
Compare
Member

atlantis plan

atlantis plan

Ran Plan for 0 projects:

Ran Plan for 0 projects:
tentype requested changes 2026-09-21 00:28:17 +00:00
Dismissed
tentype left a comment
Please [link your accounts in Keycloak](https://idp.scottylabs.org/realms/scottylabs/account/account-security/linked-accounts)
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
atlantis/plan 6/6 projects failed to plan.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: google_groups Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: keycloak Plan failed.
kennel/build skipped: no services or static sites defined
e918f5ac2d

Ran Plan for 6 projects:

  1. project: google_groups dir: tofu/google_groups workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: slack dir: tofu/slack workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: keycloak dir: tofu/keycloak workspace: default

1. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
tentype approved these changes 2026-09-21 01:11:26 +00:00
Owner

Consolidated into #353. Closing.

Consolidated into #353. Closing.
thesuperRL closed this pull request 2026-09-22 18:18:16 +00:00
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: google_groups Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan 6/6 projects failed to plan.
kennel/build skipped: no services or static sites defined

Pull request closed

Sign in to join this conversation.
No description provided.