add mangomochi7 to cmueats #278
No reviewers
Labels
No labels
needs/linking
needs/docs
needs/research
needs/upstream
type/bug
type/external
type/feature
type/refactor
urgency
high
urgency
immediate
urgency
low
urgency
medium
No milestone
No project
No assignees
3 participants
Notifications
Due date
No due date set.
Dependencies
No dependencies set
Reference
ScottyLabs/governance!278
Loading…
Reference in a new issue
No description provided.
Delete branch "mangomochi7/governance:main"
Deleting a branch is permanent. Although the deleted branch may continue to exist for a short time before it actually gets removed, it CANNOT be undone in most cases. Continue?
@indigo0445 confirmed
Ran Plan for 6 projects:
slackdir:tofu/slackworkspace:defaultgoogle_playdir:tofu/google_playworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgitdir:tofu/gitworkspace:defaultkeycloakdir:tofu/keycloakworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #274. To continue, delete the lock from #274 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
google_playdir:tofu/google_playworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #274. To continue, delete the lock from #274 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #274. To continue, delete the lock from #274 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #274. To continue, delete the lock from #274 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #274. To continue, delete the lock from #274 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.6. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #274. To continue, delete the lock from #274 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.Plan Summary
6 projects, 0 with changes, 0 with no changes, 6 failed
@thesuperRL oops this is ur id. nvm you can close this, as Ive included all new members in #308
Pull request closed