feat: add audreyha to cbp #275
No reviewers
Labels
No labels
needs/linking
needs/docs
needs/research
needs/upstream
type/bug
type/external
type/feature
type/refactor
urgency
high
urgency
immediate
urgency
low
urgency
medium
No milestone
No project
No assignees
4 participants
Notifications
Due date
No due date set.
Dependencies
No dependencies set
Reference
ScottyLabs/governance!275
Loading…
Reference in a new issue
No description provided.
Delete branch "audreyha/governance:audreyha-patch-1"
Deleting a branch is permanent. Although the deleted branch may continue to exist for a short time before it actually gets removed, it CANNOT be undone in most cases. Continue?
Your cmu.dev account is not linked, please link it here!
Once that's done, let me know and I'll approve. Thanks!
0ab7a5ed4512c59d60f6@ -6,6 +6,7 @@ members = ["xboxbedrock","liamwalk","NikhilR","audreyha"Please add a comma at the end of this line!
Ran Plan for 6 projects:
slackdir:tofu/slackworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultgoogle_playdir:tofu/google_playworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.6. project:
google_playdir:tofu/google_playworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #325. To continue, delete the lock from #325 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.Plan Summary
6 projects, 0 with changes, 0 with no changes, 6 failed
Consolidated into #353. Closing.
Pull request closed