feat: add pollyg to the CMU maps #254

Merged
xboxbedrock merged 2 commits from pollyg/governance:main into main 2026-09-21 03:41:21 +00:00
Member
No description provided.
feat: add pollyg to the CMU maps
Some checks are pending
CI / check (pull_request) Blocked by required conditions
CI / check-1 (pull_request) Blocked by required conditions
CI / build (pull_request) Blocked by required conditions
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 9m33s
edcbc2964d
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: git Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan 6/6 projects failed to plan.
kennel/build skipped: no services or static sites defined
23e838ed62

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: git dir: tofu/git workspace: default
  3. project: keycloak dir: tofu/keycloak workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: google_play dir: tofu/google_play workspace: default

1. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #224. To continue, delete the lock from #224 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
siddharthr approved these changes 2026-09-20 19:04:04 +00:00
xboxbedrock force-pushed main from 23e838ed62
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: git Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan 6/6 projects failed to plan.
kennel/build skipped: no services or static sites defined
to 47200ede32
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: google_play Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan 6/6 projects failed to plan.
kennel/build skipped: no services or static sites defined
atlantis/plan: git Plan failed.
atlantis/plan: keycloak Plan failed.
2026-09-21 03:37:03 +00:00
Compare

Ran Plan for 6 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: git dir: tofu/git workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default
  6. project: slack dir: tofu/slack workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Plan Error

Show Output
running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu init -input=false -upgrade' in '/var/lib/atlantis/repos/ScottyLabs/governance/254/default/tofu/google_play': exit status 1
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.35s


OpenTofu encountered problems during initialization, including problems
with the configuration, described below.

The OpenTofu configuration must be valid before initialization so that
OpenTofu can determine which modules and providers need to be installed.
╷
│ Error: Unclosed object
│ 
│   on google_play.tf.json line 1:
│    1: {
│ 
│ No closing brace was found for this JSON object.
╵

╷
│ Error: Root value must be object
│ 
│   on google_play.tf.json line 1:
│    1: {
│ 
│ The root value in a JSON-based configuration must be either a JSON object
│ or a JSON array of objects.
╵

╷
│ Error: Invalid start of value
│ 
│   on google_play.tf.json line 152:
│  152: <<<<<<< HEAD
│ 
│ A JSON value must start with a brace, a bracket, a number, a string, or a
│ keyword.
╵

╷
│ Error: Invalid object property name
│ 
│   on google_play.tf.json line 152:
│  152: <<<<<<< HEAD
│ 
│ A JSON object property name must be a string
╵

╷
│ Error: Missing attribute seperator comma
│ 
│   on google_play.tf.json line 152:
│  152: <<<<<<< HEAD
│ 
│ A comma must appear between each property definition in an object.
╵

╷
│ Error: Unclosed object
│ 
│   on identities.tf.json line 1:
│    1: {
│ 
│ No closing brace was found for this JSON object.
╵

╷
│ Error: Root value must be object
│ 
│   on identities.tf.json line 1:
│    1: {
│ 
│ The root value in a JSON-based configuration must be either a JSON object
│ or a JSON array of objects.
╵

╷
│ Error: Invalid start of value
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A JSON value must start with a brace, a bracket, a number, a string, or a
│ keyword.
╵

╷
│ Error: Invalid object property name
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A JSON object property name must be a string
╵

╷
│ Error: Missing attribute seperator comma
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A comma must appear between each property definition in an object.
╵



2. project: discord dir: tofu/discord workspace: default

Plan Error

Show Output
running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu init -input=false -upgrade' in '/var/lib/atlantis/repos/ScottyLabs/governance/254/default/tofu/discord': exit status 1
    Blocking waiting for file lock on package cache
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/254/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/254/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/254/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 7.40s


OpenTofu encountered problems during initialization, including problems
with the configuration, described below.

The OpenTofu configuration must be valid before initialization so that
OpenTofu can determine which modules and providers need to be installed.
╷
│ Error: Unclosed object
│ 
│   on discord.tf.json line 1:
│    1: {
│ 
│ No closing brace was found for this JSON object.
╵

╷
│ Error: Root value must be object
│ 
│   on discord.tf.json line 1:
│    1: {
│ 
│ The root value in a JSON-based configuration must be either a JSON object
│ or a JSON array of objects.
╵

╷
│ Error: Invalid start of value
│ 
│   on discord.tf.json line 861:
│  861: <<<<<<< HEAD
│ 
│ A JSON value must start with a brace, a bracket, a number, a string, or a
│ keyword.
╵

╷
│ Error: Invalid object property name
│ 
│   on discord.tf.json line 861:
│  861: <<<<<<< HEAD
│ 
│ A JSON object property name must be a string
╵

╷
│ Error: Missing attribute seperator comma
│ 
│   on discord.tf.json line 861:
│  861: <<<<<<< HEAD
│ 
│ A comma must appear between each property definition in an object.
╵

╷
│ Error: Unclosed object
│ 
│   on identities.tf.json line 1:
│    1: {
│ 
│ No closing brace was found for this JSON object.
╵

╷
│ Error: Root value must be object
│ 
│   on identities.tf.json line 1:
│    1: {
│ 
│ The root value in a JSON-based configuration must be either a JSON object
│ or a JSON array of objects.
╵

╷
│ Error: Invalid start of value
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A JSON value must start with a brace, a bracket, a number, a string, or a
│ keyword.
╵

╷
│ Error: Invalid object property name
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A JSON object property name must be a string
╵

╷
│ Error: Missing attribute seperator comma
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A comma must appear between each property definition in an object.
╵



3. project: google_groups dir: tofu/google_groups workspace: default

Plan Error

Show Output
running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu init -input=false -upgrade' in '/var/lib/atlantis/repos/ScottyLabs/governance/254/default/tofu/google_groups': exit status 1
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.40s


OpenTofu encountered problems during initialization, including problems
with the configuration, described below.

The OpenTofu configuration must be valid before initialization so that
OpenTofu can determine which modules and providers need to be installed.
╷
│ Error: Unclosed object
│ 
│   on google_groups.tf.json line 1:
│    1: {
│ 
│ No closing brace was found for this JSON object.
╵

╷
│ Error: Root value must be object
│ 
│   on google_groups.tf.json line 1:
│    1: {
│ 
│ The root value in a JSON-based configuration must be either a JSON object
│ or a JSON array of objects.
╵

╷
│ Error: Invalid start of value
│ 
│   on google_groups.tf.json line 846:
│  846: <<<<<<< HEAD
│ 
│ A JSON value must start with a brace, a bracket, a number, a string, or a
│ keyword.
╵

╷
│ Error: Invalid object property name
│ 
│   on google_groups.tf.json line 846:
│  846: <<<<<<< HEAD
│ 
│ A JSON object property name must be a string
╵

╷
│ Error: Missing attribute seperator comma
│ 
│   on google_groups.tf.json line 846:
│  846: <<<<<<< HEAD
│ 
│ A comma must appear between each property definition in an object.
╵

╷
│ Error: Unclosed object
│ 
│   on identities.tf.json line 1:
│    1: {
│ 
│ No closing brace was found for this JSON object.
╵

╷
│ Error: Root value must be object
│ 
│   on identities.tf.json line 1:
│    1: {
│ 
│ The root value in a JSON-based configuration must be either a JSON object
│ or a JSON array of objects.
╵

╷
│ Error: Invalid start of value
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A JSON value must start with a brace, a bracket, a number, a string, or a
│ keyword.
╵

╷
│ Error: Invalid object property name
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A JSON object property name must be a string
╵

╷
│ Error: Missing attribute seperator comma
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A comma must appear between each property definition in an object.
╵



4. project: git dir: tofu/git workspace: default

Plan Error

Show Output
running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu init -input=false -upgrade' in '/var/lib/atlantis/repos/ScottyLabs/governance/254/default/tofu/git': exit status 1
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.38s


OpenTofu encountered problems during initialization, including problems
with the configuration, described below.

The OpenTofu configuration must be valid before initialization so that
OpenTofu can determine which modules and providers need to be installed.
╷
│ Error: Unclosed object
│ 
│   on forgejo_memberships.tf.json line 1:
│    1: {
│ 
│ No closing brace was found for this JSON object.
╵

╷
│ Error: Root value must be object
│ 
│   on forgejo_memberships.tf.json line 1:
│    1: {
│ 
│ The root value in a JSON-based configuration must be either a JSON object
│ or a JSON array of objects.
╵

╷
│ Error: Invalid start of value
│ 
│   on forgejo_memberships.tf.json line 220:
│  220: <<<<<<< HEAD
│ 
│ A JSON value must start with a brace, a bracket, a number, a string, or a
│ keyword.
╵

╷
│ Error: Invalid object property name
│ 
│   on forgejo_memberships.tf.json line 220:
│  220: <<<<<<< HEAD
│ 
│ A JSON object property name must be a string
╵

╷
│ Error: Missing attribute seperator comma
│ 
│   on forgejo_memberships.tf.json line 220:
│  220: <<<<<<< HEAD
│ 
│ A comma must appear between each property definition in an object.
╵

╷
│ Error: Unclosed object
│ 
│   on github_memberships.tf.json line 1:
│    1: {
│ 
│ No closing brace was found for this JSON object.
╵

╷
│ Error: Root value must be object
│ 
│   on github_memberships.tf.json line 1:
│    1: {
│ 
│ The root value in a JSON-based configuration must be either a JSON object
│ or a JSON array of objects.
╵

╷
│ Error: Invalid start of value
│ 
│   on github_memberships.tf.json line 239:
│  239: <<<<<<< HEAD
│ 
│ A JSON value must start with a brace, a bracket, a number, a string, or a
│ keyword.
╵

╷
│ Error: Invalid object property name
│ 
│   on github_memberships.tf.json line 239:
│  239: <<<<<<< HEAD
│ 
│ A JSON object property name must be a string
╵

╷
│ Error: Missing attribute seperator comma
│ 
│   on github_memberships.tf.json line 239:
│  239: <<<<<<< HEAD
│ 
│ A comma must appear between each property definition in an object.
╵

╷
│ Error: Unclosed object
│ 
│   on identities.tf.json line 1:
│    1: {
│ 
│ No closing brace was found for this JSON object.
╵

╷
│ Error: Root value must be object
│ 
│   on identities.tf.json line 1:
│    1: {
│ 
│ The root value in a JSON-based configuration must be either a JSON object
│ or a JSON array of objects.
╵

╷
│ Error: Invalid start of value
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A JSON value must start with a brace, a bracket, a number, a string, or a
│ keyword.
╵

╷
│ Error: Invalid object property name
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A JSON object property name must be a string
╵

╷
│ Error: Missing attribute seperator comma
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A comma must appear between each property definition in an object.
╵



5. project: keycloak dir: tofu/keycloak workspace: default

Plan Error

Show Output
running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu init -input=false -upgrade' in '/var/lib/atlantis/repos/ScottyLabs/governance/254/default/tofu/keycloak': exit status 1
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.41s


OpenTofu encountered problems during initialization, including problems
with the configuration, described below.

The OpenTofu configuration must be valid before initialization so that
OpenTofu can determine which modules and providers need to be installed.
╷
│ Error: Unclosed object
│ 
│   on identities.tf.json line 1:
│    1: {
│ 
│ No closing brace was found for this JSON object.
╵

╷
│ Error: Root value must be object
│ 
│   on identities.tf.json line 1:
│    1: {
│ 
│ The root value in a JSON-based configuration must be either a JSON object
│ or a JSON array of objects.
╵

╷
│ Error: Invalid start of value
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A JSON value must start with a brace, a bracket, a number, a string, or a
│ keyword.
╵

╷
│ Error: Invalid object property name
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A JSON object property name must be a string
╵

╷
│ Error: Missing attribute seperator comma
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A comma must appear between each property definition in an object.
╵

╷
│ Error: Unclosed object
│ 
│   on keycloak_memberships.tf.json line 1:
│    1: {
│ 
│ No closing brace was found for this JSON object.
╵

╷
│ Error: Root value must be object
│ 
│   on keycloak_memberships.tf.json line 1:
│    1: {
│ 
│ The root value in a JSON-based configuration must be either a JSON object
│ or a JSON array of objects.
╵

╷
│ Error: Unclosed object
│ 
│   on keycloak_memberships.tf.json line 2:
│    2:   "resource": {
│ 
│ No closing brace was found for this JSON object.
╵

╷
│ Error: Unclosed object
│ 
│   on keycloak_memberships.tf.json line 3:
│    3:     "keycloak_group_memberships": {
│ 
│ No closing brace was found for this JSON object.
╵

╷
│ Error: Unclosed object
│ 
│   on keycloak_memberships.tf.json line 251:
│  251:       "project_cmu-maps_maps": {
│ 
│ No closing brace was found for this JSON object.
╵

╷
│ Error: Invalid start of value
│ 
│   on keycloak_memberships.tf.json line 256:
│  256: <<<<<<< HEAD
│ 
│ A JSON value must start with a brace, a bracket, a number, a string, or a
│ keyword.
╵

╷
│ Error: Missing attribute seperator comma
│ 
│   on keycloak_memberships.tf.json line 256:
│  256: <<<<<<< HEAD
│ 
│ A comma must appear between each value in an array.
╵



6. project: slack dir: tofu/slack workspace: default

Plan Error

Show Output
running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu init -input=false -upgrade' in '/var/lib/atlantis/repos/ScottyLabs/governance/254/default/tofu/slack': exit status 1
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.40s


OpenTofu encountered problems during initialization, including problems
with the configuration, described below.

The OpenTofu configuration must be valid before initialization so that
OpenTofu can determine which modules and providers need to be installed.
╷
│ Error: Unclosed object
│ 
│   on identities.tf.json line 1:
│    1: {
│ 
│ No closing brace was found for this JSON object.
╵

╷
│ Error: Root value must be object
│ 
│   on identities.tf.json line 1:
│    1: {
│ 
│ The root value in a JSON-based configuration must be either a JSON object
│ or a JSON array of objects.
╵

╷
│ Error: Invalid start of value
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A JSON value must start with a brace, a bracket, a number, a string, or a
│ keyword.
╵

╷
│ Error: Invalid object property name
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A JSON object property name must be a string
╵

╷
│ Error: Missing attribute seperator comma
│ 
│   on identities.tf.json line 587:
│  587: <<<<<<< HEAD
│ 
│ A comma must appear between each property definition in an object.
╵

╷
│ Error: Unclosed object
│ 
│   on slack.tf.json line 1:
│    1: {
│ 
│ No closing brace was found for this JSON object.
╵

╷
│ Error: Root value must be object
│ 
│   on slack.tf.json line 1:
│    1: {
│ 
│ The root value in a JSON-based configuration must be either a JSON object
│ or a JSON array of objects.
╵

╷
│ Error: Invalid start of value
│ 
│   on slack.tf.json line 612:
│  612: <<<<<<< HEAD
│ 
│ A JSON value must start with a brace, a bracket, a number, a string, or a
│ keyword.
╵

╷
│ Error: Invalid object property name
│ 
│   on slack.tf.json line 612:
│  612: <<<<<<< HEAD
│ 
│ A JSON object property name must be a string
╵

╷
│ Error: Missing attribute seperator comma
│ 
│   on slack.tf.json line 612:
│  612: <<<<<<< HEAD
│ 
│ A comma must appear between each property definition in an object.
╵



Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Error** <details><summary>Show Output</summary> ``` running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu init -input=false -upgrade' in '/var/lib/atlantis/repos/ScottyLabs/governance/254/default/tofu/google_play': exit status 1 Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.35s OpenTofu encountered problems during initialization, including problems with the configuration, described below. The OpenTofu configuration must be valid before initialization so that OpenTofu can determine which modules and providers need to be installed. ╷ │ Error: Unclosed object │ │ on google_play.tf.json line 1: │ 1: { │ │ No closing brace was found for this JSON object. ╵ ╷ │ Error: Root value must be object │ │ on google_play.tf.json line 1: │ 1: { │ │ The root value in a JSON-based configuration must be either a JSON object │ or a JSON array of objects. ╵ ╷ │ Error: Invalid start of value │ │ on google_play.tf.json line 152: │ 152: <<<<<<< HEAD │ │ A JSON value must start with a brace, a bracket, a number, a string, or a │ keyword. ╵ ╷ │ Error: Invalid object property name │ │ on google_play.tf.json line 152: │ 152: <<<<<<< HEAD │ │ A JSON object property name must be a string ╵ ╷ │ Error: Missing attribute seperator comma │ │ on google_play.tf.json line 152: │ 152: <<<<<<< HEAD │ │ A comma must appear between each property definition in an object. ╵ ╷ │ Error: Unclosed object │ │ on identities.tf.json line 1: │ 1: { │ │ No closing brace was found for this JSON object. ╵ ╷ │ Error: Root value must be object │ │ on identities.tf.json line 1: │ 1: { │ │ The root value in a JSON-based configuration must be either a JSON object │ or a JSON array of objects. ╵ ╷ │ Error: Invalid start of value │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A JSON value must start with a brace, a bracket, a number, a string, or a │ keyword. ╵ ╷ │ Error: Invalid object property name │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A JSON object property name must be a string ╵ ╷ │ Error: Missing attribute seperator comma │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A comma must appear between each property definition in an object. ╵ ``` </details> --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Error** <details><summary>Show Output</summary> ``` running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu init -input=false -upgrade' in '/var/lib/atlantis/repos/ScottyLabs/governance/254/default/tofu/discord': exit status 1 Blocking waiting for file lock on package cache Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/254/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/254/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/254/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 7.40s OpenTofu encountered problems during initialization, including problems with the configuration, described below. The OpenTofu configuration must be valid before initialization so that OpenTofu can determine which modules and providers need to be installed. ╷ │ Error: Unclosed object │ │ on discord.tf.json line 1: │ 1: { │ │ No closing brace was found for this JSON object. ╵ ╷ │ Error: Root value must be object │ │ on discord.tf.json line 1: │ 1: { │ │ The root value in a JSON-based configuration must be either a JSON object │ or a JSON array of objects. ╵ ╷ │ Error: Invalid start of value │ │ on discord.tf.json line 861: │ 861: <<<<<<< HEAD │ │ A JSON value must start with a brace, a bracket, a number, a string, or a │ keyword. ╵ ╷ │ Error: Invalid object property name │ │ on discord.tf.json line 861: │ 861: <<<<<<< HEAD │ │ A JSON object property name must be a string ╵ ╷ │ Error: Missing attribute seperator comma │ │ on discord.tf.json line 861: │ 861: <<<<<<< HEAD │ │ A comma must appear between each property definition in an object. ╵ ╷ │ Error: Unclosed object │ │ on identities.tf.json line 1: │ 1: { │ │ No closing brace was found for this JSON object. ╵ ╷ │ Error: Root value must be object │ │ on identities.tf.json line 1: │ 1: { │ │ The root value in a JSON-based configuration must be either a JSON object │ or a JSON array of objects. ╵ ╷ │ Error: Invalid start of value │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A JSON value must start with a brace, a bracket, a number, a string, or a │ keyword. ╵ ╷ │ Error: Invalid object property name │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A JSON object property name must be a string ╵ ╷ │ Error: Missing attribute seperator comma │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A comma must appear between each property definition in an object. ╵ ``` </details> --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Error** <details><summary>Show Output</summary> ``` running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu init -input=false -upgrade' in '/var/lib/atlantis/repos/ScottyLabs/governance/254/default/tofu/google_groups': exit status 1 Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.40s OpenTofu encountered problems during initialization, including problems with the configuration, described below. The OpenTofu configuration must be valid before initialization so that OpenTofu can determine which modules and providers need to be installed. ╷ │ Error: Unclosed object │ │ on google_groups.tf.json line 1: │ 1: { │ │ No closing brace was found for this JSON object. ╵ ╷ │ Error: Root value must be object │ │ on google_groups.tf.json line 1: │ 1: { │ │ The root value in a JSON-based configuration must be either a JSON object │ or a JSON array of objects. ╵ ╷ │ Error: Invalid start of value │ │ on google_groups.tf.json line 846: │ 846: <<<<<<< HEAD │ │ A JSON value must start with a brace, a bracket, a number, a string, or a │ keyword. ╵ ╷ │ Error: Invalid object property name │ │ on google_groups.tf.json line 846: │ 846: <<<<<<< HEAD │ │ A JSON object property name must be a string ╵ ╷ │ Error: Missing attribute seperator comma │ │ on google_groups.tf.json line 846: │ 846: <<<<<<< HEAD │ │ A comma must appear between each property definition in an object. ╵ ╷ │ Error: Unclosed object │ │ on identities.tf.json line 1: │ 1: { │ │ No closing brace was found for this JSON object. ╵ ╷ │ Error: Root value must be object │ │ on identities.tf.json line 1: │ 1: { │ │ The root value in a JSON-based configuration must be either a JSON object │ or a JSON array of objects. ╵ ╷ │ Error: Invalid start of value │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A JSON value must start with a brace, a bracket, a number, a string, or a │ keyword. ╵ ╷ │ Error: Invalid object property name │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A JSON object property name must be a string ╵ ╷ │ Error: Missing attribute seperator comma │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A comma must appear between each property definition in an object. ╵ ``` </details> --- ### 4. project: `git` dir: `tofu/git` workspace: `default` **Plan Error** <details><summary>Show Output</summary> ``` running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu init -input=false -upgrade' in '/var/lib/atlantis/repos/ScottyLabs/governance/254/default/tofu/git': exit status 1 Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.38s OpenTofu encountered problems during initialization, including problems with the configuration, described below. The OpenTofu configuration must be valid before initialization so that OpenTofu can determine which modules and providers need to be installed. ╷ │ Error: Unclosed object │ │ on forgejo_memberships.tf.json line 1: │ 1: { │ │ No closing brace was found for this JSON object. ╵ ╷ │ Error: Root value must be object │ │ on forgejo_memberships.tf.json line 1: │ 1: { │ │ The root value in a JSON-based configuration must be either a JSON object │ or a JSON array of objects. ╵ ╷ │ Error: Invalid start of value │ │ on forgejo_memberships.tf.json line 220: │ 220: <<<<<<< HEAD │ │ A JSON value must start with a brace, a bracket, a number, a string, or a │ keyword. ╵ ╷ │ Error: Invalid object property name │ │ on forgejo_memberships.tf.json line 220: │ 220: <<<<<<< HEAD │ │ A JSON object property name must be a string ╵ ╷ │ Error: Missing attribute seperator comma │ │ on forgejo_memberships.tf.json line 220: │ 220: <<<<<<< HEAD │ │ A comma must appear between each property definition in an object. ╵ ╷ │ Error: Unclosed object │ │ on github_memberships.tf.json line 1: │ 1: { │ │ No closing brace was found for this JSON object. ╵ ╷ │ Error: Root value must be object │ │ on github_memberships.tf.json line 1: │ 1: { │ │ The root value in a JSON-based configuration must be either a JSON object │ or a JSON array of objects. ╵ ╷ │ Error: Invalid start of value │ │ on github_memberships.tf.json line 239: │ 239: <<<<<<< HEAD │ │ A JSON value must start with a brace, a bracket, a number, a string, or a │ keyword. ╵ ╷ │ Error: Invalid object property name │ │ on github_memberships.tf.json line 239: │ 239: <<<<<<< HEAD │ │ A JSON object property name must be a string ╵ ╷ │ Error: Missing attribute seperator comma │ │ on github_memberships.tf.json line 239: │ 239: <<<<<<< HEAD │ │ A comma must appear between each property definition in an object. ╵ ╷ │ Error: Unclosed object │ │ on identities.tf.json line 1: │ 1: { │ │ No closing brace was found for this JSON object. ╵ ╷ │ Error: Root value must be object │ │ on identities.tf.json line 1: │ 1: { │ │ The root value in a JSON-based configuration must be either a JSON object │ or a JSON array of objects. ╵ ╷ │ Error: Invalid start of value │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A JSON value must start with a brace, a bracket, a number, a string, or a │ keyword. ╵ ╷ │ Error: Invalid object property name │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A JSON object property name must be a string ╵ ╷ │ Error: Missing attribute seperator comma │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A comma must appear between each property definition in an object. ╵ ``` </details> --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Error** <details><summary>Show Output</summary> ``` running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu init -input=false -upgrade' in '/var/lib/atlantis/repos/ScottyLabs/governance/254/default/tofu/keycloak': exit status 1 Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.41s OpenTofu encountered problems during initialization, including problems with the configuration, described below. The OpenTofu configuration must be valid before initialization so that OpenTofu can determine which modules and providers need to be installed. ╷ │ Error: Unclosed object │ │ on identities.tf.json line 1: │ 1: { │ │ No closing brace was found for this JSON object. ╵ ╷ │ Error: Root value must be object │ │ on identities.tf.json line 1: │ 1: { │ │ The root value in a JSON-based configuration must be either a JSON object │ or a JSON array of objects. ╵ ╷ │ Error: Invalid start of value │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A JSON value must start with a brace, a bracket, a number, a string, or a │ keyword. ╵ ╷ │ Error: Invalid object property name │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A JSON object property name must be a string ╵ ╷ │ Error: Missing attribute seperator comma │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A comma must appear between each property definition in an object. ╵ ╷ │ Error: Unclosed object │ │ on keycloak_memberships.tf.json line 1: │ 1: { │ │ No closing brace was found for this JSON object. ╵ ╷ │ Error: Root value must be object │ │ on keycloak_memberships.tf.json line 1: │ 1: { │ │ The root value in a JSON-based configuration must be either a JSON object │ or a JSON array of objects. ╵ ╷ │ Error: Unclosed object │ │ on keycloak_memberships.tf.json line 2: │ 2: "resource": { │ │ No closing brace was found for this JSON object. ╵ ╷ │ Error: Unclosed object │ │ on keycloak_memberships.tf.json line 3: │ 3: "keycloak_group_memberships": { │ │ No closing brace was found for this JSON object. ╵ ╷ │ Error: Unclosed object │ │ on keycloak_memberships.tf.json line 251: │ 251: "project_cmu-maps_maps": { │ │ No closing brace was found for this JSON object. ╵ ╷ │ Error: Invalid start of value │ │ on keycloak_memberships.tf.json line 256: │ 256: <<<<<<< HEAD │ │ A JSON value must start with a brace, a bracket, a number, a string, or a │ keyword. ╵ ╷ │ Error: Missing attribute seperator comma │ │ on keycloak_memberships.tf.json line 256: │ 256: <<<<<<< HEAD │ │ A comma must appear between each value in an array. ╵ ``` </details> --- ### 6. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Error** <details><summary>Show Output</summary> ``` running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu init -input=false -upgrade' in '/var/lib/atlantis/repos/ScottyLabs/governance/254/default/tofu/slack': exit status 1 Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.40s OpenTofu encountered problems during initialization, including problems with the configuration, described below. The OpenTofu configuration must be valid before initialization so that OpenTofu can determine which modules and providers need to be installed. ╷ │ Error: Unclosed object │ │ on identities.tf.json line 1: │ 1: { │ │ No closing brace was found for this JSON object. ╵ ╷ │ Error: Root value must be object │ │ on identities.tf.json line 1: │ 1: { │ │ The root value in a JSON-based configuration must be either a JSON object │ or a JSON array of objects. ╵ ╷ │ Error: Invalid start of value │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A JSON value must start with a brace, a bracket, a number, a string, or a │ keyword. ╵ ╷ │ Error: Invalid object property name │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A JSON object property name must be a string ╵ ╷ │ Error: Missing attribute seperator comma │ │ on identities.tf.json line 587: │ 587: <<<<<<< HEAD │ │ A comma must appear between each property definition in an object. ╵ ╷ │ Error: Unclosed object │ │ on slack.tf.json line 1: │ 1: { │ │ No closing brace was found for this JSON object. ╵ ╷ │ Error: Root value must be object │ │ on slack.tf.json line 1: │ 1: { │ │ The root value in a JSON-based configuration must be either a JSON object │ or a JSON array of objects. ╵ ╷ │ Error: Invalid start of value │ │ on slack.tf.json line 612: │ 612: <<<<<<< HEAD │ │ A JSON value must start with a brace, a bracket, a number, a string, or a │ keyword. ╵ ╷ │ Error: Invalid object property name │ │ on slack.tf.json line 612: │ 612: <<<<<<< HEAD │ │ A JSON object property name must be a string ╵ ╷ │ Error: Missing attribute seperator comma │ │ on slack.tf.json line 612: │ 612: <<<<<<< HEAD │ │ A comma must appear between each property definition in an object. ╵ ``` </details> --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
xboxbedrock force-pushed main from 47200ede32
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: google_play Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan 6/6 projects failed to plan.
kennel/build skipped: no services or static sites defined
atlantis/plan: git Plan failed.
atlantis/plan: keycloak Plan failed.
to 36e6b2232f
All checks were successful
atlantis/plan: slack Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_play Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 2 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
2026-09-21 03:37:38 +00:00
Compare

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.13s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-maps_ch0_pollyg will be created
+ resource "null_resource" "slack_cmu-maps_ch0_pollyg" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C05T7QRN4LF"
          + "user"    = "U0ALF27AC2Y"
        }
    }

  # null_resource.slack_hub_pollyg will be created
+ resource "null_resource" "slack_hub_pollyg" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0ALF27AC2Y"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 0 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # googleplay_app_iam.maps_pollyg will be created
+ resource "googleplay_app_iam" "maps_pollyg" {
+       app_id               = "com.cmumaps.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "pollyg@andrew.cmu.edu"
    }

  # googleplay_user.pollyg will be created
+ resource "googleplay_user" "pollyg" {
+       email                = "pollyg@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 2 to add, 0 to change, 0 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.25s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_pollyg will be created
+ resource "google_cloud_identity_group_membership" "tech_pollyg" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "pollyg@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.19s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-maps_maps will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-maps_maps" {
        id       = "scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64"
!       members  = [
+           "pollyg",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "pollyg",
            # (72 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 2 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 2 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.19s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.pollyg will be created
+ resource "discord_member_roles" "pollyg" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "895126440956297307"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1526003694708195448"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # forgejo_team_member.cmu-maps_pollyg will be created
+ resource "forgejo_team_member" "cmu-maps_pollyg" {
+       team_id = 13
+       user    = "pollyg"
    }

  # github_team_membership.cmu-maps_pollyg will be created
+ resource "github_team_membership" "cmu-maps_pollyg" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "10872360"
+       username = "symphonyno25yu"
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 0 to change, 0 to destroy.


Plan Summary

6 projects, 6 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.13s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_cmu-maps_ch0_pollyg will be created + resource "null_resource" "slack_cmu-maps_ch0_pollyg" { + id = (known after apply) + triggers = { + "channel" = "C05T7QRN4LF" + "user" = "U0ALF27AC2Y" } } # null_resource.slack_hub_pollyg will be created + resource "null_resource" "slack_hub_pollyg" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0ALF27AC2Y" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # googleplay_app_iam.maps_pollyg will be created + resource "googleplay_app_iam" "maps_pollyg" { + app_id = "com.cmumaps.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "pollyg@andrew.cmu.edu" } # googleplay_user.pollyg will be created + resource "googleplay_user" "pollyg" { + email = "pollyg@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.25s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_pollyg will be created + resource "google_cloud_identity_group_membership" "tech_pollyg" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "pollyg@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.19s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-maps_maps will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-maps_maps" { id = "scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64" ! members = [ + "pollyg", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "pollyg", # (72 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 2 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 2 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.19s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.pollyg will be created + resource "discord_member_roles" "pollyg" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "895126440956297307" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1526003694708195448" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # forgejo_team_member.cmu-maps_pollyg will be created + resource "forgejo_team_member" "cmu-maps_pollyg" { + team_id = 13 + user = "pollyg" } # github_team_membership.cmu-maps_pollyg will be created + resource "github_team_membership" "cmu-maps_pollyg" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "10872360" + username = "symphonyno25yu" } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### Plan Summary 6 projects, 6 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis apply

atlantis apply

Ran Apply for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: google_play dir: tofu/google_play workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: keycloak dir: tofu/keycloak workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.20s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_hub_pollyg: Creating...
null_resource.slack_cmu-maps_ch0_pollyg: Creating...
null_resource.slack_hub_pollyg: Provisioning with 'local-exec'...
null_resource.slack_hub_pollyg (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0ALF27AC2Y"]
null_resource.slack_cmu-maps_ch0_pollyg: Provisioning with 'local-exec'...
null_resource.slack_cmu-maps_ch0_pollyg (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C05T7QRN4LF --user U0ALF27AC2Y"]
null_resource.slack_cmu-maps_ch0_pollyg: Creation complete after 0s [id=4812679365331161717]
null_resource.slack_hub_pollyg: Creation complete after 0s [id=3236858972933711000]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

2. project: discord dir: tofu/discord workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.06s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.pollyg: Creating...
discord_member_roles.pollyg: Creation complete after 1s [id=1353118209024331948:895126440956297307]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

3. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.07s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
googleplay_user.pollyg: Creating...
googleplay_user.pollyg: Creation complete after 1s [id=pollyg@andrew.cmu.edu]
googleplay_app_iam.maps_pollyg: Creating...
googleplay_app_iam.maps_pollyg: Creation complete after 1s

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_pollyg: Creating...
google_cloud_identity_group_membership.tech_pollyg: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/102743546800389632234]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.19s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
github_team_membership.cmu-maps_pollyg: Creating...
forgejo_team_member.cmu-maps_pollyg: Creating...
forgejo_team_member.cmu-maps_pollyg: Creation complete after 0s
github_team_membership.cmu-maps_pollyg: Creation complete after 1s [id=10872360:symphonyno25yu]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

6. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cmu-maps_maps: Modifying... [id=scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_cmu-maps_maps: Modifications complete after 1s [id=scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64]
keycloak_group_memberships.projects: Modifications complete after 2s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 2 changed, 0 destroyed.

Apply Summary

6 projects, 6 successful, 0 failed, 0 errored

Ran Apply for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.20s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_hub_pollyg: Creating... null_resource.slack_cmu-maps_ch0_pollyg: Creating... null_resource.slack_hub_pollyg: Provisioning with 'local-exec'... null_resource.slack_hub_pollyg (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0ALF27AC2Y"] null_resource.slack_cmu-maps_ch0_pollyg: Provisioning with 'local-exec'... null_resource.slack_cmu-maps_ch0_pollyg (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C05T7QRN4LF --user U0ALF27AC2Y"] null_resource.slack_cmu-maps_ch0_pollyg: Creation complete after 0s [id=4812679365331161717] null_resource.slack_hub_pollyg: Creation complete after 0s [id=3236858972933711000] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.06s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.pollyg: Creating... discord_member_roles.pollyg: Creation complete after 1s [id=1353118209024331948:895126440956297307] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.07s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ googleplay_user.pollyg: Creating... googleplay_user.pollyg: Creation complete after 1s [id=pollyg@andrew.cmu.edu] googleplay_app_iam.maps_pollyg: Creating... googleplay_app_iam.maps_pollyg: Creation complete after 1s Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_pollyg: Creating... google_cloud_identity_group_membership.tech_pollyg: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/102743546800389632234] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.19s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ github_team_membership.cmu-maps_pollyg: Creating... forgejo_team_member.cmu-maps_pollyg: Creating... forgejo_team_member.cmu-maps_pollyg: Creation complete after 0s github_team_membership.cmu-maps_pollyg: Creation complete after 1s [id=10872360:symphonyno25yu] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 6. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cmu-maps_maps: Modifying... [id=scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_cmu-maps_maps: Modifications complete after 1s [id=scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64] keycloak_group_memberships.projects: Modifications complete after 2s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 2 changed, 0 destroyed. ``` </details> --- ### Apply Summary 6 projects, 6 successful, 0 failed, 0 errored

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
All checks were successful
atlantis/plan: slack Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_play Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 2 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
Some workflows are waiting to be reviewed.
Sign in to join this conversation.
No description provided.