feat: Add sgli to courses #241

Merged
xboxbedrock merged 3 commits from sgli/governance:main into main 2026-09-19 22:08:10 +00:00
Member
No description provided.
Update data/teams/cmu-courses.toml
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
Validate / Governance (pull_request_target) Successful in 6m8s
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
CI / build (pull_request) Has been cancelled
c4dbe830ef
etashjha changed title from Update data/teams/cmu-courses.toml to feat: Add sgli to courses 2026-09-19 20:19:24 +00:00
etashjha approved these changes 2026-09-19 20:19:27 +00:00
Dismissed
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: slack Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan 6/6 projects failed to plan.
ac8da96d7c
scottylabs-bot dismissed etashjha's review 2026-09-19 20:23:16 +00:00
Reason:

New commits pushed, approval review dismissed automatically according to repository settings

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: keycloak dir: tofu/keycloak workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: discord dir: tofu/discord workspace: default
  5. project: google_play dir: tofu/google_play workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
etashjha approved these changes 2026-09-19 21:19:35 +00:00
xboxbedrock force-pushed main from ac8da96d7c
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: slack Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan 6/6 projects failed to plan.
to 373fa2e747
All checks were successful
atlantis/apply 6/6 projects applied successfully.
CI / build (pull_request) Has been skipped
atlantis/plan: slack Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_play Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 3 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
atlantis/plan: git Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
Validate / Governance (pull_request_target) Successful in 7m44s
CI / check-1 (pull_request) Successful in 13m27s
CI / check (pull_request) Successful in 0s
2026-09-19 22:06:39 +00:00
Compare

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.27s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-courses_ch0_sgli will be created
+ resource "null_resource" "slack_cmu-courses_ch0_sgli" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0150RGAG1L"
          + "user"    = "U0C3XQ4EV24"
        }
    }

  # null_resource.slack_hub_sgli will be created
+ resource "null_resource" "slack_hub_sgli" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0C3XQ4EV24"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 0 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.28s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # googleplay_app_iam.courses_sgli will be created
+ resource "googleplay_app_iam" "courses_sgli" {
+       app_id               = "courses.cmu.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "sgli@andrew.cmu.edu"
    }

  # googleplay_user.sgli will be created
+ resource "googleplay_user" "sgli" {
+       email                = "sgli@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 2 to add, 0 to change, 0 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.26s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_sgli will be created
+ resource "google_cloud_identity_group_membership" "tech_sgli" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "sgli@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.29s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-courses will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-courses" {
        id       = "scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77"
!       members  = [
+           "sgli",
            # (6 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-courses_courses will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-courses_courses" {
        id       = "scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406"
!       members  = [
+           "sgli",
            # (6 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "sgli",
            # (49 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 3 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 3 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.30s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.sgli will be created
+ resource "discord_member_roles" "sgli" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "529745895911325727"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1500016758700904488"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.24s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # forgejo_team_member.cmu-courses_sgli will be created
+ resource "forgejo_team_member" "cmu-courses_sgli" {
+       team_id = 12
+       user    = "sgli"
    }

  # github_team_membership.cmu-courses_sgli will be created
+ resource "github_team_membership" "cmu-courses_sgli" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "3643907"
+       username = "duallyhazard"
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 0 to change, 0 to destroy.


Plan Summary

6 projects, 6 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.27s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_cmu-courses_ch0_sgli will be created + resource "null_resource" "slack_cmu-courses_ch0_sgli" { + id = (known after apply) + triggers = { + "channel" = "C0150RGAG1L" + "user" = "U0C3XQ4EV24" } } # null_resource.slack_hub_sgli will be created + resource "null_resource" "slack_hub_sgli" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0C3XQ4EV24" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.28s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # googleplay_app_iam.courses_sgli will be created + resource "googleplay_app_iam" "courses_sgli" { + app_id = "courses.cmu.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "sgli@andrew.cmu.edu" } # googleplay_user.sgli will be created + resource "googleplay_user" "sgli" { + email = "sgli@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.26s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_sgli will be created + resource "google_cloud_identity_group_membership" "tech_sgli" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "sgli@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.29s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-courses will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-courses" { id = "scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77" ! members = [ + "sgli", # (6 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-courses_courses will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-courses_courses" { id = "scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406" ! members = [ + "sgli", # (6 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "sgli", # (49 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 3 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 3 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.30s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.sgli will be created + resource "discord_member_roles" "sgli" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "529745895911325727" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1500016758700904488" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.24s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # forgejo_team_member.cmu-courses_sgli will be created + resource "forgejo_team_member" "cmu-courses_sgli" { + team_id = 12 + user = "sgli" } # github_team_membership.cmu-courses_sgli will be created + resource "github_team_membership" "cmu-courses_sgli" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "3643907" + username = "duallyhazard" } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### Plan Summary 6 projects, 6 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis apply

atlantis apply

Ran Apply for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: google_play dir: tofu/google_play workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: keycloak dir: tofu/keycloak workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.15s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_cmu-courses_ch0_sgli: Creating...
null_resource.slack_hub_sgli: Creating...
null_resource.slack_cmu-courses_ch0_sgli: Provisioning with 'local-exec'...
null_resource.slack_cmu-courses_ch0_sgli (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0150RGAG1L --user U0C3XQ4EV24"]
null_resource.slack_hub_sgli: Provisioning with 'local-exec'...
null_resource.slack_hub_sgli (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C3XQ4EV24"]
null_resource.slack_hub_sgli: Creation complete after 0s [id=681108518133013991]
null_resource.slack_cmu-courses_ch0_sgli: Creation complete after 0s [id=2318002929926355193]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

2. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.12s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.sgli: Creating...
discord_member_roles.sgli: Creation complete after 1s [id=1353118209024331948:529745895911325727]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.13s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_sgli: Creating...
google_cloud_identity_group_membership.tech_sgli: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/100148624715954529112]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

4. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
googleplay_user.sgli: Creating...
googleplay_user.sgli: Creation complete after 1s [id=sgli@andrew.cmu.edu]
googleplay_app_iam.courses_sgli: Creating...
googleplay_app_iam.courses_sgli: Creation complete after 1s

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.23s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.cmu-courses_sgli: Creating...
forgejo_team_member.cmu-courses_sgli: Creation complete after 0s
github_team_membership.cmu-courses_sgli: Creating...
github_team_membership.cmu-courses_sgli: Creation complete after 1s [id=3643907:duallyhazard]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

6. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cmu-courses_courses: Modifying... [id=scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406]
keycloak_group_memberships.project_cmu-courses: Modifying... [id=scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_cmu-courses_courses: Modifications complete after 1s [id=scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406]
keycloak_group_memberships.project_cmu-courses: Modifications complete after 2s [id=scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77]
keycloak_group_memberships.projects: Modifications complete after 2s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 3 changed, 0 destroyed.

Apply Summary

6 projects, 6 successful, 0 failed, 0 errored

Ran Apply for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.15s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_cmu-courses_ch0_sgli: Creating... null_resource.slack_hub_sgli: Creating... null_resource.slack_cmu-courses_ch0_sgli: Provisioning with 'local-exec'... null_resource.slack_cmu-courses_ch0_sgli (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0150RGAG1L --user U0C3XQ4EV24"] null_resource.slack_hub_sgli: Provisioning with 'local-exec'... null_resource.slack_hub_sgli (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C3XQ4EV24"] null_resource.slack_hub_sgli: Creation complete after 0s [id=681108518133013991] null_resource.slack_cmu-courses_ch0_sgli: Creation complete after 0s [id=2318002929926355193] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.12s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.sgli: Creating... discord_member_roles.sgli: Creation complete after 1s [id=1353118209024331948:529745895911325727] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.13s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_sgli: Creating... google_cloud_identity_group_membership.tech_sgli: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/100148624715954529112] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ googleplay_user.sgli: Creating... googleplay_user.sgli: Creation complete after 1s [id=sgli@andrew.cmu.edu] googleplay_app_iam.courses_sgli: Creating... googleplay_app_iam.courses_sgli: Creation complete after 1s Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.23s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.cmu-courses_sgli: Creating... forgejo_team_member.cmu-courses_sgli: Creation complete after 0s github_team_membership.cmu-courses_sgli: Creating... github_team_membership.cmu-courses_sgli: Creation complete after 1s [id=3643907:duallyhazard] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 6. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cmu-courses_courses: Modifying... [id=scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406] keycloak_group_memberships.project_cmu-courses: Modifying... [id=scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_cmu-courses_courses: Modifications complete after 1s [id=scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406] keycloak_group_memberships.project_cmu-courses: Modifications complete after 2s [id=scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77] keycloak_group_memberships.projects: Modifications complete after 2s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 3 changed, 0 destroyed. ``` </details> --- ### Apply Summary 6 projects, 6 successful, 0 failed, 0 errored
xboxbedrock referenced this pull request from a commit 2026-09-19 22:08:11 +00:00

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
brightz referenced this pull request from a commit 2026-09-26 22:03:28 +00:00
Sign in to join this conversation.
No description provided.