feat: add kennethz to cmu-maps #228

Merged
xboxbedrock merged 2 commits from kennethz/governance:main into main 2026-09-21 03:07:47 +00:00
Member

Add Kenneth Zhong to cmu-maps toml member data.

Add Kenneth Zhong to cmu-maps toml member data.
feat: add kennethz to cmu-maps
All checks were successful
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 11m54s
CI / build (pull_request) Has been skipped
CI / check-1 (pull_request) Successful in 6m1s
CI / check (pull_request) Successful in 0s
fca567bd64
siddharthr approved these changes 2026-09-19 18:53:48 +00:00
Dismissed
siddharthr approved these changes 2026-09-19 18:54:17 +00:00
Dismissed
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: google_play Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan 6/6 projects failed to plan.
kennel/build skipped: no services or static sites defined
03f83a5a85
scottylabs-bot dismissed siddharthr's review 2026-09-19 19:26:49 +00:00
Reason:

New commits pushed, approval review dismissed automatically according to repository settings

Ran Plan for 6 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: keycloak dir: tofu/keycloak workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: git dir: tofu/git workspace: default
  5. project: google_groups dir: tofu/google_groups workspace: default
  6. project: slack dir: tofu/slack workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #234. To continue, delete the lock from #234 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
siddharthr approved these changes 2026-09-20 19:04:56 +00:00
xboxbedrock force-pushed main from 03f83a5a85
Some checks failed
atlantis/plan: slack Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan 6/6 projects failed to plan.
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: google_play Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: keycloak Plan failed.
atlantis/plan: discord Plan failed.
to 439ba02029
All checks were successful
atlantis/plan: slack Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: google_play Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 2 to change, 0 to destroy.
atlantis/plan: git Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
2026-09-21 03:04:57 +00:00
Compare

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.26s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-maps_ch0_kennethz will be created
+ resource "null_resource" "slack_cmu-maps_ch0_kennethz" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C05T7QRN4LF"
          + "user"    = "U0BTMURLEFL"
        }
    }

  # null_resource.slack_hub_kennethz will be created
+ resource "null_resource" "slack_hub_kennethz" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0BTMURLEFL"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 0 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/228/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/228/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/228/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 7.29s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # googleplay_app_iam.maps_kennethz will be created
+ resource "googleplay_app_iam" "maps_kennethz" {
+       app_id               = "com.cmumaps.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "kzhong2@andrew.cmu.edu"
    }

  # googleplay_user.kennethz will be created
+ resource "googleplay_user" "kennethz" {
+       email                = "kzhong2@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 2 to add, 0 to change, 0 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.27s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_kennethz will be created
+ resource "google_cloud_identity_group_membership" "tech_kennethz" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "kzhong2@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.30s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-maps_maps will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-maps_maps" {
        id       = "scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64"
!       members  = [
+           "kzhong2",
            # (3 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "kzhong2",
            # (70 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 2 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 2 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.29s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.kennethz will be created
+ resource "discord_member_roles" "kennethz" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "1535829582588940318"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1526003694708195448"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 7.31s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # forgejo_team_member.cmu-maps_kennethz will be created
+ resource "forgejo_team_member" "cmu-maps_kennethz" {
+       team_id = 13
+       user    = "kzhong2"
    }

  # github_team_membership.cmu-maps_kennethz will be created
+ resource "github_team_membership" "cmu-maps_kennethz" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "10872360"
+       username = "amazingkenneth"
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 0 to change, 0 to destroy.


Plan Summary

6 projects, 6 with changes, 0 with no changes, 0 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.26s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_cmu-maps_ch0_kennethz will be created + resource "null_resource" "slack_cmu-maps_ch0_kennethz" { + id = (known after apply) + triggers = { + "channel" = "C05T7QRN4LF" + "user" = "U0BTMURLEFL" } } # null_resource.slack_hub_kennethz will be created + resource "null_resource" "slack_hub_kennethz" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0BTMURLEFL" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/228/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/228/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/228/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 7.29s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # googleplay_app_iam.maps_kennethz will be created + resource "googleplay_app_iam" "maps_kennethz" { + app_id = "com.cmumaps.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "kzhong2@andrew.cmu.edu" } # googleplay_user.kennethz will be created + resource "googleplay_user" "kennethz" { + email = "kzhong2@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.27s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_kennethz will be created + resource "google_cloud_identity_group_membership" "tech_kennethz" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "kzhong2@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.30s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-maps_maps will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-maps_maps" { id = "scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64" ! members = [ + "kzhong2", # (3 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "kzhong2", # (70 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 2 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 2 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.29s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.kennethz will be created + resource "discord_member_roles" "kennethz" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "1535829582588940318" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1526003694708195448" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 7.31s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # forgejo_team_member.cmu-maps_kennethz will be created + resource "forgejo_team_member" "cmu-maps_kennethz" { + team_id = 13 + user = "kzhong2" } # github_team_membership.cmu-maps_kennethz will be created + resource "github_team_membership" "cmu-maps_kennethz" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "10872360" + username = "amazingkenneth" } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### Plan Summary 6 projects, 6 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Member

atlantis apply

atlantis apply

Ran Apply for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: google_play dir: tofu/google_play workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: keycloak dir: tofu/keycloak workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.17s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_hub_kennethz: Creating...
null_resource.slack_cmu-maps_ch0_kennethz: Creating...
null_resource.slack_hub_kennethz: Provisioning with 'local-exec'...
null_resource.slack_hub_kennethz (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0BTMURLEFL"]
null_resource.slack_cmu-maps_ch0_kennethz: Provisioning with 'local-exec'...
null_resource.slack_cmu-maps_ch0_kennethz (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C05T7QRN4LF --user U0BTMURLEFL"]
null_resource.slack_hub_kennethz: Creation complete after 0s [id=2506502594701195393]
null_resource.slack_cmu-maps_ch0_kennethz: Creation complete after 0s [id=4137649222352534344]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

2. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.21s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.kennethz: Creating...
discord_member_roles.kennethz: Creation complete after 1s [id=1353118209024331948:1535829582588940318]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.27s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_kennethz: Creating...
google_cloud_identity_group_membership.tech_kennethz: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/117935351361858038339]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

4. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.23s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
googleplay_user.kennethz: Creating...
googleplay_user.kennethz: Creation complete after 1s [id=kzhong2@andrew.cmu.edu]
googleplay_app_iam.maps_kennethz: Creating...
googleplay_app_iam.maps_kennethz: Creation complete after 1s

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
github_team_membership.cmu-maps_kennethz: Creating...
forgejo_team_member.cmu-maps_kennethz: Creating...
forgejo_team_member.cmu-maps_kennethz: Creation complete after 0s
github_team_membership.cmu-maps_kennethz: Creation complete after 2s [id=10872360:amazingkenneth]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

6. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.15s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cmu-maps_maps: Modifying... [id=scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_cmu-maps_maps: Modifications complete after 2s [id=scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64]
keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 2 changed, 0 destroyed.

Apply Summary

6 projects, 6 successful, 0 failed, 0 errored

Ran Apply for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.17s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_hub_kennethz: Creating... null_resource.slack_cmu-maps_ch0_kennethz: Creating... null_resource.slack_hub_kennethz: Provisioning with 'local-exec'... null_resource.slack_hub_kennethz (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0BTMURLEFL"] null_resource.slack_cmu-maps_ch0_kennethz: Provisioning with 'local-exec'... null_resource.slack_cmu-maps_ch0_kennethz (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C05T7QRN4LF --user U0BTMURLEFL"] null_resource.slack_hub_kennethz: Creation complete after 0s [id=2506502594701195393] null_resource.slack_cmu-maps_ch0_kennethz: Creation complete after 0s [id=4137649222352534344] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.21s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.kennethz: Creating... discord_member_roles.kennethz: Creation complete after 1s [id=1353118209024331948:1535829582588940318] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.27s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_kennethz: Creating... google_cloud_identity_group_membership.tech_kennethz: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/117935351361858038339] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.23s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ googleplay_user.kennethz: Creating... googleplay_user.kennethz: Creation complete after 1s [id=kzhong2@andrew.cmu.edu] googleplay_app_iam.maps_kennethz: Creating... googleplay_app_iam.maps_kennethz: Creation complete after 1s Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ github_team_membership.cmu-maps_kennethz: Creating... forgejo_team_member.cmu-maps_kennethz: Creating... forgejo_team_member.cmu-maps_kennethz: Creation complete after 0s github_team_membership.cmu-maps_kennethz: Creation complete after 2s [id=10872360:amazingkenneth] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 6. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.15s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cmu-maps_maps: Modifying... [id=scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_cmu-maps_maps: Modifications complete after 2s [id=scottylabs/group-memberships/b50ccd92-08af-4a27-b275-2929ccfd4f64] keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 2 changed, 0 destroyed. ``` </details> --- ### Apply Summary 6 projects, 6 successful, 0 failed, 0 errored

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
brightz referenced this pull request from a commit 2026-09-26 22:03:28 +00:00
Sign in to join this conversation.
No description provided.