feat: create cmu-study.toml #215

Merged
anish merged 2 commits from annadavi/governance:main into main 2026-09-19 00:21:01 +00:00
Member

New team cmu-study.toml

New team cmu-study.toml
create cmu-study.toml
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
CI / build (pull_request) Has been cancelled
Validate / Governance (pull_request_target) Has been cancelled
b7552fa978
anish changed title from create cmu-study.toml to feat: create cmu-study.toml 2026-09-18 23:16:54 +00:00
anish approved these changes 2026-09-18 23:17:26 +00:00
anish force-pushed main from b7552fa978
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
CI / build (pull_request) Has been cancelled
Validate / Governance (pull_request_target) Has been cancelled
to 69e8efb499
Some checks failed
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
CI / check (pull_request) Has been cancelled
Validate / Governance (pull_request_target) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
2026-09-18 23:17:32 +00:00
Compare
anish force-pushed main from 69e8efb499
Some checks failed
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
CI / check (pull_request) Has been cancelled
Validate / Governance (pull_request_target) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
to 96db1e1044
All checks were successful
Validate / Governance (pull_request_target) Successful in 9m4s
CI / check-1 (pull_request) Successful in 12m58s
CI / check (pull_request) Successful in 0s
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
2026-09-18 23:23:12 +00:00
Compare
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 3 to add, 0 to change, 0 to destroy.
atlantis/plan: openbao Plan: 6 to add, 24 to change, 0 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: posthog Plan: 3 to add, 0 to change, 0 to destroy.
atlantis/plan: matrix_bridges Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: sentry Plan: 4 to add, 0 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 38 to add, 1 to change, 0 to destroy.
atlantis/plan: discord Plan: 3 to add, 0 to change, 0 to destroy.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: git Plan failed.
atlantis/plan 1/10 projects failed to plan.
8c41fd4672

Ran Plan for 10 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: openbao dir: tofu/openbao workspace: default
  3. project: google_play dir: tofu/google_play workspace: default
  4. project: matrix_bridges dir: tofu/matrix_bridges workspace: default
  5. project: posthog dir: tofu/posthog workspace: default
  6. project: sentry dir: tofu/sentry workspace: default
  7. project: google_groups dir: tofu/google_groups workspace: default
  8. project: keycloak dir: tofu/keycloak workspace: default
  9. project: discord dir: tofu/discord workspace: default
  10. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 18.98s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-study_ch0_annadavi will be created
+ resource "null_resource" "slack_cmu-study_ch0_annadavi" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C09KN6M3P53"
          + "user"    = "U07M24AKRGV"
        }
    }

  # null_resource.slack_hub_annadavi will be created
+ resource "null_resource" "slack_hub_annadavi" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U07M24AKRGV"
        }
    }

  # null_resource.slack_leads_annadavi will be created
+ resource "null_resource" "slack_leads_annadavi" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0AN1M0HDV0"
          + "user"    = "U07M24AKRGV"
        }
    }

Plan: 3 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 3 to add, 0 to change, 0 to destroy.


2. project: openbao dir: tofu/openbao workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
 Downloading crates ...
  Downloaded rustls-webpki v0.103.15
  Downloaded rustls v0.23.45
    Blocking waiting for file lock on package cache
   Compiling rustls v0.23.45
   Compiling rustls-webpki v0.103.15
   Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/215/default/crates/governance-schema)
   Compiling ureq v3.3.0
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/215/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/215/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/215/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 18.81s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # vault_identity_group.bus-sign_admins will be updated in-place
! resource "vault_identity_group" "bus-sign_admins" {
        id                         = "1b72d8b5-8511-fe92-c98c-2ddb8cbf17a1"
        name                       = "bus-sign-admins"
!       policies                   = [
-           "bus-sign-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.cal_admins will be updated in-place
! resource "vault_identity_group" "cal_admins" {
        id                         = "da96ea82-8537-f0a1-0d42-a7ab1cc3c006"
        name                       = "cal-admins"
!       policies                   = [
-           "cal-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.changelog_admins will be updated in-place
! resource "vault_identity_group" "changelog_admins" {
        id                         = "d73c0ef4-a7d2-d171-ebd1-1c15cee174c8"
        name                       = "changelog-admins"
!       policies                   = [
-           "changelog-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.cmugpt-agent_admins will be updated in-place
! resource "vault_identity_group" "cmugpt-agent_admins" {
        id                         = "cb081cd2-a3cd-81e1-8c20-0914cac5c213"
        name                       = "cmugpt-agent-admins"
!       policies                   = [
-           "cmugpt-agent-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.cmugpt-mcp-server_admins will be updated in-place
! resource "vault_identity_group" "cmugpt-mcp-server_admins" {
        id                         = "a4717a6f-015f-9fae-c6a2-76657b8a7464"
        name                       = "cmugpt-mcp-server-admins"
!       policies                   = [
-           "cmugpt-mcp-server-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.cmugpt-sms-surface_admins will be updated in-place
! resource "vault_identity_group" "cmugpt-sms-surface_admins" {
        id                         = "780f9498-dee7-8cac-9bbc-e1feab5ea2fb"
        name                       = "cmugpt-sms-surface-admins"
!       policies                   = [
-           "cmugpt-sms-surface-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.cmugpt-surface_admins will be updated in-place
! resource "vault_identity_group" "cmugpt-surface_admins" {
        id                         = "3ae18ef3-f370-f163-767a-accfb092ea0d"
        name                       = "cmugpt-surface-admins"
!       policies                   = [
-           "cmugpt-surface-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.components_admins will be updated in-place
! resource "vault_identity_group" "components_admins" {
        id                         = "d2971366-64a4-984e-1e4e-dad3cd07c841"
        name                       = "components-admins"
!       policies                   = [
-           "components-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.courses_admins will be updated in-place
! resource "vault_identity_group" "courses_admins" {
        id                         = "d090ae17-1992-86a5-1f8b-092c937092e8"
        name                       = "courses-admins"
!       policies                   = [
-           "courses-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.dalmatian_admins will be updated in-place
! resource "vault_identity_group" "dalmatian_admins" {
        id                         = "b1d88b18-8b44-c2ea-9220-fc53ee7e3c7e"
        name                       = "dalmatian-admins"
!       policies                   = [
-           "dalmatian-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.discord-verify_admins will be updated in-place
! resource "vault_identity_group" "discord-verify_admins" {
        id                         = "c3e29b3b-0ce5-61a3-3a69-e950f297fa78"
        name                       = "discord-verify-admins"
!       policies                   = [
-           "discord-verify-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.documentation_admins will be updated in-place
! resource "vault_identity_group" "documentation_admins" {
        id                         = "35c5df83-0e17-31fc-7b82-3365d5526dd5"
        name                       = "documentation-admins"
!       policies                   = [
-           "documentation-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.eats_admins will be updated in-place
! resource "vault_identity_group" "eats_admins" {
        id                         = "05959815-c896-9703-32eb-7ba84604303c"
        name                       = "eats-admins"
!       policies                   = [
-           "eats-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.groupme-mirror_admins will be updated in-place
! resource "vault_identity_group" "groupme-mirror_admins" {
        id                         = "e264aabf-8766-67fd-f0bf-860a240ee2c2"
        name                       = "groupme-mirror-admins"
!       policies                   = [
-           "groupme-mirror-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.housing_admins will be updated in-place
! resource "vault_identity_group" "housing_admins" {
        id                         = "7c688f49-59f1-1593-1e19-16f70b86eae2"
        name                       = "housing-admins"
!       policies                   = [
-           "housing-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.insta_admins will be updated in-place
! resource "vault_identity_group" "insta_admins" {
        id                         = "be24e607-e8d5-9fa3-a11a-9a8ca05ede04"
        name                       = "insta-admins"
!       policies                   = [
-           "insta-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.link-shortener_admins will be updated in-place
! resource "vault_identity_group" "link-shortener_admins" {
        id                         = "d3fa8791-4bd4-db20-68b6-79db5ac36a25"
        name                       = "link-shortener-admins"
!       policies                   = [
-           "link-shortener-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.lost-and-found-v2_admins will be updated in-place
! resource "vault_identity_group" "lost-and-found-v2_admins" {
        id                         = "db8b5072-bb86-c9d0-e93f-951c2901bd15"
        name                       = "lost-and-found-v2-admins"
!       policies                   = [
-           "lost-and-found-v2-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.maps_admins will be updated in-place
! resource "vault_identity_group" "maps_admins" {
        id                         = "6b2bd85e-e311-ae83-8a07-30100f0ac12f"
        name                       = "maps-admins"
!       policies                   = [
-           "maps-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.occ-main_admins will be updated in-place
! resource "vault_identity_group" "occ-main_admins" {
        id                         = "4d40db70-1bac-505e-ec8d-5a9d5cf15ac7"
        name                       = "occ-main-admins"
!       policies                   = [
-           "occ-main-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.quest_admins will be updated in-place
! resource "vault_identity_group" "quest_admins" {
        id                         = "c03f2eab-1701-f315-fc17-39814ddcb34b"
        name                       = "quest-admins"
!       policies                   = [
-           "quest-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.scottylabs-org_admins will be updated in-place
! resource "vault_identity_group" "scottylabs-org_admins" {
        id                         = "0dbd914d-b96c-de7b-1ec5-a0ef330b20e0"
        name                       = "scottylabs-org-admins"
!       policies                   = [
-           "scottylabs-org-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.study_admins will be created
+ resource "vault_identity_group" "study_admins" {
+       external_member_entity_ids = false
+       external_member_group_ids  = false
+       external_policies          = false
+       id                         = (known after apply)
+       name                       = "study-admins"
+       policies                   = [
+           "study-prod",
        ]
+       type                       = "external"
    }

  # vault_identity_group.study_members will be created
+ resource "vault_identity_group" "study_members" {
+       external_member_entity_ids = false
+       external_member_group_ids  = false
+       external_policies          = false
+       id                         = (known after apply)
+       name                       = "study-members"
+       policies                   = [
+           "study-dev",
        ]
+       type                       = "external"
    }

  # vault_identity_group.tartan-vote_admins will be updated in-place
! resource "vault_identity_group" "tartan-vote_admins" {
        id                         = "ff77ad01-20b1-c320-37bd-6185b3f98d80"
        name                       = "tartan-vote-admins"
!       policies                   = [
-           "tartan-vote-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.terrier_admins will be updated in-place
! resource "vault_identity_group" "terrier_admins" {
        id                         = "956692a1-84f6-c95c-6d40-2382ead13476"
        name                       = "terrier-admins"
!       policies                   = [
-           "terrier-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group_alias.study_admins will be created
+ resource "vault_identity_group_alias" "study_admins" {
+       canonical_id   = (known after apply)
+       id             = (known after apply)
+       mount_accessor = "auth_oidc_375ac979"
+       name           = "/projects/study/admins"
    }

  # vault_identity_group_alias.study_members will be created
+ resource "vault_identity_group_alias" "study_members" {
+       canonical_id   = (known after apply)
+       id             = (known after apply)
+       mount_accessor = "auth_oidc_375ac979"
+       name           = "/projects/study"
    }

  # vault_policy.study_dev will be created
+ resource "vault_policy" "study_dev" {
+       allow_overwrite = (known after apply)
+       id              = (known after apply)
+       name            = "study-dev"
+       policy          = <<-EOT
            path "secret/data/secretspec/study/dev/*" {
              capabilities = ["create", "read", "update", "delete", "list"]
            }
            
            path "secret/metadata/secretspec/study/dev/*" {
              capabilities = ["list", "read", "delete"]
            }
            
            path "secret/data/secretspec/study/preview/*" {
              capabilities = ["create", "read", "update", "delete", "list"]
            }
            
            path "secret/metadata/secretspec/study/preview/*" {
              capabilities = ["list", "read", "delete"]
            }
        EOT
    }

  # vault_policy.study_prod will be created
+ resource "vault_policy" "study_prod" {
+       allow_overwrite = (known after apply)
+       id              = (known after apply)
+       name            = "study-prod"
+       policy          = <<-EOT
            path "secret/data/secretspec/study/prod/*" {
              capabilities = ["create", "read", "update", "delete", "list"]
            }
            
            path "secret/metadata/secretspec/study/prod/*" {
              capabilities = ["list", "read", "delete"]
            }
            
            path "secret/data/secretspec/study/staging/*" {
              capabilities = ["create", "read", "update", "delete", "list"]
            }
            
            path "secret/metadata/secretspec/study/staging/*" {
              capabilities = ["list", "read", "delete"]
            }
            
            path "secret/data/secretspec/study/ci/*" {
              capabilities = ["create", "read", "update", "delete", "list"]
            }
            
            path "secret/metadata/secretspec/study/ci/*" {
              capabilities = ["list", "read", "delete"]
            }
        EOT
    }

Plan: 6 to add, 24 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p openbao
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p openbao
    

Plan: 6 to add, 24 to change, 0 to destroy.


3. project: google_play dir: tofu/google_play workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 18.80s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

4. project: matrix_bridges dir: tofu/matrix_bridges workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 18.90s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_bridge_login_cmu_study will be created
+ resource "null_resource" "slack_bridge_login_cmu_study" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C09KN6M3P53"
        }
    }

  # synapse_mautrix_slack_link.cmu_study will be created
+ resource "synapse_mautrix_slack_link" "cmu_study" {
+       discord_channel_id       = "1512270354943578132"
+       discord_relay_configured = (known after apply)
+       id                       = (known after apply)
+       room_alias               = (known after apply)
+       room_id                  = (known after apply)
+       slack_bridged            = (known after apply)
+       slack_channel_id         = "C09KN6M3P53"
+       team_name                = "CMU Study"
+       team_slug                = "cmu-study"
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p matrix_bridges
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p matrix_bridges
    

Plan: 2 to add, 0 to change, 0 to destroy.


5. project: posthog dir: tofu/posthog workspace: default

Show Output
Updating crates.io index
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 18.93s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # posthog_project.cmu-study_study will be created
+ resource "posthog_project" "cmu-study_study" {
+       api_token       = (sensitive value)
+       id              = (known after apply)
+       name            = "study"
+       organization_id = (known after apply)
+       timezone        = (known after apply)
    }

  # vault_kv_secret_v2.cmu-study_study_posthog_host will be created
+ resource "vault_kv_secret_v2" "cmu-study_study_posthog_host" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/POSTHOG_HOST"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.cmu-study_study_posthog_key will be created
+ resource "vault_kv_secret_v2" "cmu-study_study_posthog_key" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/POSTHOG_KEY"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

Plan: 3 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p posthog
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p posthog
    

Plan: 3 to add, 0 to change, 0 to destroy.


6. project: sentry dir: tofu/sentry workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 18.98s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # sentry_key.cmu-study_study will be created
+ resource "sentry_key" "cmu-study_study" {
+       dsn                      = (sensitive value)
+       dsn_csp                  = (known after apply)
+       dsn_public               = (known after apply)
+       dsn_secret               = (sensitive value)
+       id                       = (known after apply)
+       javascript_loader_script = (known after apply)
+       name                     = "default"
+       organization             = "scottylabs"
+       project                  = "study"
+       project_id               = (known after apply)
+       public                   = (known after apply)
+       rate_limit_count         = (known after apply)
+       rate_limit_window        = (known after apply)
+       secret                   = (sensitive value)
    }

  # sentry_project.cmu-study_study will be created
+ resource "sentry_project" "cmu-study_study" {
+       client_security       = (known after apply)
+       digests_max_delay     = (known after apply)
+       digests_min_delay     = (known after apply)
+       features              = (known after apply)
+       filters               = (known after apply)
+       fingerprinting_rules  = (known after apply)
+       grouping_enhancements = (known after apply)
+       highlight_tags        = (known after apply)
+       id                    = (known after apply)
+       internal_id           = (known after apply)
+       name                  = "study"
+       organization          = "scottylabs"
+       resolve_age           = (known after apply)
+       slug                  = "study"
+       teams                 = [
+           "cmu-study",
        ]
    }

  # sentry_team.cmu-study will be created
+ resource "sentry_team" "cmu-study" {
+       has_access   = (known after apply)
+       id           = (known after apply)
+       internal_id  = (known after apply)
+       is_member    = (known after apply)
+       is_pending   = (known after apply)
+       name         = "CMU Study"
+       organization = "scottylabs"
+       slug         = "cmu-study"
+       team_id      = (known after apply)
    }

  # vault_kv_secret_v2.cmu-study_study_sentry_dsn will be created
+ resource "vault_kv_secret_v2" "cmu-study_study_sentry_dsn" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/SENTRY_DSN"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

Plan: 4 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p sentry
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p sentry
    

Plan: 4 to add, 0 to change, 0 to destroy.


7. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 18.98s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.ops_annadavi will be created
+ resource "google_cloud_identity_group_membership" "ops_annadavi" {
+       create_time = (known after apply)
+       group       = "groups/03fwokq036zsp1f"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "annadavi@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

  # google_cloud_identity_group_membership.tech_annadavi will be created
+ resource "google_cloud_identity_group_membership" "tech_annadavi" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "annadavi@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 2 to add, 0 to change, 0 to destroy.


8. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 18.94s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group.project_cmu-study_study will be created
+ resource "keycloak_group" "project_cmu-study_study" {
+       id        = (known after apply)
+       name      = "study"
+       parent_id = "9e6890c8-50a0-4b82-86b9-91791fffac25"
+       path      = (known after apply)
+       realm_id  = "scottylabs"
    }

  # keycloak_group.project_cmu-study_study_admins will be created
+ resource "keycloak_group" "project_cmu-study_study_admins" {
+       id        = (known after apply)
+       name      = "admins"
+       parent_id = (known after apply)
+       path      = (known after apply)
+       realm_id  = "scottylabs"
    }

  # keycloak_group_memberships.project_cmu-study_study will be created
+ resource "keycloak_group_memberships" "project_cmu-study_study" {
+       group_id = (known after apply)
+       id       = (known after apply)
+       members  = [
+           "annadavi",
        ]
+       realm_id = "scottylabs"
    }

  # keycloak_group_memberships.project_cmu-study_study_admins will be created
+ resource "keycloak_group_memberships" "project_cmu-study_study_admins" {
+       group_id = (known after apply)
+       id       = (known after apply)
+       members  = [
+           "annadavi",
        ]
+       realm_id = "scottylabs"
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "annadavi",
            # (35 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_openid_client.study will be created
+ resource "keycloak_openid_client" "study" {
+       access_token_lifespan                     = (known after apply)
+       access_type                               = "CONFIDENTIAL"
+       always_display_in_console                 = false
+       backchannel_logout_session_required       = true
+       client_authenticator_type                 = "client-secret"
+       client_id                                 = "study"
+       client_offline_session_idle_timeout       = (known after apply)
+       client_offline_session_max_lifespan       = (known after apply)
+       client_secret                             = (sensitive value)
+       client_secret_wo                          = (write-only attribute)
+       client_session_idle_timeout               = (known after apply)
+       client_session_max_lifespan               = (known after apply)
+       consent_required                          = (known after apply)
+       direct_access_grants_enabled              = false
+       display_on_consent_screen                 = (known after apply)
+       enabled                                   = true
+       exclude_issuer_from_auth_response         = (known after apply)
+       exclude_session_state_from_auth_response  = (known after apply)
+       frontchannel_logout_enabled               = (known after apply)
+       full_scope_allowed                        = true
+       id                                        = (known after apply)
+       implicit_flow_enabled                     = (known after apply)
+       import                                    = false
+       name                                      = "study"
+       oauth2_device_authorization_grant_enabled = false
+       oauth2_jwt_authorization_grant_enabled    = false
+       realm_id                                  = "scottylabs"
+       require_dpop_bound_tokens                 = (known after apply)
+       resource_server_id                        = (known after apply)
+       service_account_user_id                   = (known after apply)
+       service_accounts_enabled                  = (known after apply)
+       standard_flow_enabled                     = true
+       standard_token_exchange_enabled           = false
+       use_refresh_tokens                        = true
+       use_refresh_tokens_client_credentials     = false
+       valid_post_logout_redirect_uris           = (known after apply)
+       valid_redirect_uris                       = [
+           "https://oauth.scottylabs.org/oauth2/callback",
        ]
+       web_origins                               = [
+           "+",
        ]
    }

  # keycloak_openid_client.study_dev will be created
+ resource "keycloak_openid_client" "study_dev" {
+       access_token_lifespan                     = (known after apply)
+       access_type                               = "CONFIDENTIAL"
+       always_display_in_console                 = false
+       backchannel_logout_session_required       = true
+       client_authenticator_type                 = "client-secret"
+       client_id                                 = "study-dev"
+       client_offline_session_idle_timeout       = (known after apply)
+       client_offline_session_max_lifespan       = (known after apply)
+       client_secret                             = (sensitive value)
+       client_secret_wo                          = (write-only attribute)
+       client_session_idle_timeout               = (known after apply)
+       client_session_max_lifespan               = (known after apply)
+       consent_required                          = (known after apply)
+       direct_access_grants_enabled              = false
+       display_on_consent_screen                 = (known after apply)
+       enabled                                   = true
+       exclude_issuer_from_auth_response         = (known after apply)
+       exclude_session_state_from_auth_response  = (known after apply)
+       frontchannel_logout_enabled               = (known after apply)
+       full_scope_allowed                        = true
+       id                                        = (known after apply)
+       implicit_flow_enabled                     = (known after apply)
+       import                                    = false
+       name                                      = "study-dev"
+       oauth2_device_authorization_grant_enabled = false
+       oauth2_jwt_authorization_grant_enabled    = false
+       realm_id                                  = "scottylabs"
+       require_dpop_bound_tokens                 = (known after apply)
+       resource_server_id                        = (known after apply)
+       service_account_user_id                   = (known after apply)
+       service_accounts_enabled                  = (known after apply)
+       standard_flow_enabled                     = true
+       standard_token_exchange_enabled           = false
+       use_refresh_tokens                        = true
+       use_refresh_tokens_client_credentials     = false
+       valid_post_logout_redirect_uris           = (known after apply)
+       valid_redirect_uris                       = [
+           "http://localhost:8090/oauth2/callback",
        ]
+       web_origins                               = [
+           "+",
        ]
    }

  # keycloak_openid_client.study_staging will be created
+ resource "keycloak_openid_client" "study_staging" {
+       access_token_lifespan                     = (known after apply)
+       access_type                               = "CONFIDENTIAL"
+       always_display_in_console                 = false
+       backchannel_logout_session_required       = true
+       client_authenticator_type                 = "client-secret"
+       client_id                                 = "study-staging"
+       client_offline_session_idle_timeout       = (known after apply)
+       client_offline_session_max_lifespan       = (known after apply)
+       client_secret                             = (sensitive value)
+       client_secret_wo                          = (write-only attribute)
+       client_session_idle_timeout               = (known after apply)
+       client_session_max_lifespan               = (known after apply)
+       consent_required                          = (known after apply)
+       direct_access_grants_enabled              = false
+       display_on_consent_screen                 = (known after apply)
+       enabled                                   = true
+       exclude_issuer_from_auth_response         = (known after apply)
+       exclude_session_state_from_auth_response  = (known after apply)
+       frontchannel_logout_enabled               = (known after apply)
+       full_scope_allowed                        = true
+       id                                        = (known after apply)
+       implicit_flow_enabled                     = (known after apply)
+       import                                    = false
+       name                                      = "study-staging"
+       oauth2_device_authorization_grant_enabled = false
+       oauth2_jwt_authorization_grant_enabled    = false
+       realm_id                                  = "scottylabs"
+       require_dpop_bound_tokens                 = (known after apply)
+       resource_server_id                        = (known after apply)
+       service_account_user_id                   = (known after apply)
+       service_accounts_enabled                  = (known after apply)
+       standard_flow_enabled                     = true
+       standard_token_exchange_enabled           = false
+       use_refresh_tokens                        = true
+       use_refresh_tokens_client_credentials     = false
+       valid_post_logout_redirect_uris           = (known after apply)
+       valid_redirect_uris                       = [
+           "https://oauth.scottylabs.org/oauth2/callback",
        ]
+       web_origins                               = [
+           "+",
        ]
    }

  # keycloak_openid_group_membership_protocol_mapper.study will be created
+ resource "keycloak_openid_group_membership_protocol_mapper" "study" {
+       add_to_access_token        = true
+       add_to_id_token            = true
+       add_to_token_introspection = false
+       add_to_userinfo            = true
+       claim_name                 = "groups"
+       client_id                  = (known after apply)
+       full_path                  = true
+       id                         = (known after apply)
+       name                       = "groups"
+       realm_id                   = "scottylabs"
    }

  # keycloak_openid_group_membership_protocol_mapper.study_dev will be created
+ resource "keycloak_openid_group_membership_protocol_mapper" "study_dev" {
+       add_to_access_token        = true
+       add_to_id_token            = true
+       add_to_token_introspection = false
+       add_to_userinfo            = true
+       claim_name                 = "groups"
+       client_id                  = (known after apply)
+       full_path                  = true
+       id                         = (known after apply)
+       name                       = "groups"
+       realm_id                   = "scottylabs"
    }

  # keycloak_openid_group_membership_protocol_mapper.study_staging will be created
+ resource "keycloak_openid_group_membership_protocol_mapper" "study_staging" {
+       add_to_access_token        = true
+       add_to_id_token            = true
+       add_to_token_introspection = false
+       add_to_userinfo            = true
+       claim_name                 = "groups"
+       client_id                  = (known after apply)
+       full_path                  = true
+       id                         = (known after apply)
+       name                       = "groups"
+       realm_id                   = "scottylabs"
    }

  # vault_kv_secret_v2.study_dev_keycloak_realm will be created
+ resource "vault_kv_secret_v2" "study_dev_keycloak_realm" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/dev/KEYCLOAK_REALM"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_dev_keycloak_url will be created
+ resource "vault_kv_secret_v2" "study_dev_keycloak_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/dev/KEYCLOAK_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_dev_oauth_relay_url will be created
+ resource "vault_kv_secret_v2" "study_dev_oauth_relay_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/dev/OAUTH_RELAY_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_dev_oidc_client_id will be created
+ resource "vault_kv_secret_v2" "study_dev_oidc_client_id" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/dev/OIDC_CLIENT_ID"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_dev_oidc_client_secret will be created
+ resource "vault_kv_secret_v2" "study_dev_oidc_client_secret" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/dev/OIDC_CLIENT_SECRET"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_dev_project_admin_group will be created
+ resource "vault_kv_secret_v2" "study_dev_project_admin_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/dev/PROJECT_ADMIN_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_dev_project_group will be created
+ resource "vault_kv_secret_v2" "study_dev_project_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/dev/PROJECT_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_preview_keycloak_realm will be created
+ resource "vault_kv_secret_v2" "study_preview_keycloak_realm" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/preview/KEYCLOAK_REALM"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_preview_keycloak_url will be created
+ resource "vault_kv_secret_v2" "study_preview_keycloak_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/preview/KEYCLOAK_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_preview_oauth_relay_url will be created
+ resource "vault_kv_secret_v2" "study_preview_oauth_relay_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/preview/OAUTH_RELAY_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_preview_oidc_client_id will be created
+ resource "vault_kv_secret_v2" "study_preview_oidc_client_id" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/preview/OIDC_CLIENT_ID"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_preview_oidc_client_secret will be created
+ resource "vault_kv_secret_v2" "study_preview_oidc_client_secret" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/preview/OIDC_CLIENT_SECRET"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_preview_project_admin_group will be created
+ resource "vault_kv_secret_v2" "study_preview_project_admin_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/preview/PROJECT_ADMIN_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_preview_project_group will be created
+ resource "vault_kv_secret_v2" "study_preview_project_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/preview/PROJECT_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_prod_keycloak_realm will be created
+ resource "vault_kv_secret_v2" "study_prod_keycloak_realm" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/KEYCLOAK_REALM"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_prod_keycloak_url will be created
+ resource "vault_kv_secret_v2" "study_prod_keycloak_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/KEYCLOAK_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_prod_oauth_relay_url will be created
+ resource "vault_kv_secret_v2" "study_prod_oauth_relay_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/OAUTH_RELAY_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_prod_oidc_client_id will be created
+ resource "vault_kv_secret_v2" "study_prod_oidc_client_id" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/OIDC_CLIENT_ID"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_prod_oidc_client_secret will be created
+ resource "vault_kv_secret_v2" "study_prod_oidc_client_secret" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/OIDC_CLIENT_SECRET"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_prod_project_admin_group will be created
+ resource "vault_kv_secret_v2" "study_prod_project_admin_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/PROJECT_ADMIN_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_prod_project_group will be created
+ resource "vault_kv_secret_v2" "study_prod_project_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/PROJECT_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_staging_keycloak_realm will be created
+ resource "vault_kv_secret_v2" "study_staging_keycloak_realm" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/staging/KEYCLOAK_REALM"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_staging_keycloak_url will be created
+ resource "vault_kv_secret_v2" "study_staging_keycloak_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/staging/KEYCLOAK_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_staging_oauth_relay_url will be created
+ resource "vault_kv_secret_v2" "study_staging_oauth_relay_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/staging/OAUTH_RELAY_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_staging_oidc_client_id will be created
+ resource "vault_kv_secret_v2" "study_staging_oidc_client_id" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/staging/OIDC_CLIENT_ID"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_staging_oidc_client_secret will be created
+ resource "vault_kv_secret_v2" "study_staging_oidc_client_secret" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/staging/OIDC_CLIENT_SECRET"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_staging_project_admin_group will be created
+ resource "vault_kv_secret_v2" "study_staging_project_admin_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/staging/PROJECT_ADMIN_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_staging_project_group will be created
+ resource "vault_kv_secret_v2" "study_staging_project_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/staging/PROJECT_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

Plan: 38 to add, 1 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 38 to add, 1 to change, 0 to destroy.


9. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 19.04s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_channel_permission.cmu-study_ch0 will be created
+ resource "discord_channel_permission" "cmu-study_ch0" {
+       allow        = 1024
+       channel_id   = "1512270354943578132"
+       id           = (known after apply)
+       overwrite_id = (known after apply)
+       type         = "role"
    }

  # discord_member_roles.annadavi will be created
+ resource "discord_member_roles" "annadavi" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "991766392590311455"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1519792320755662928"
        }
+       role {
+           has_role = true
+           role_id  = (known after apply)
        }
    }

  # discord_role.cmu-study will be created
+ resource "discord_role" "cmu-study" {
+       hoist       = false
+       id          = (known after apply)
+       managed     = (known after apply)
+       mentionable = false
+       name        = "CMU Study"
+       permissions = 0
+       position    = (known after apply)
+       server_id   = "1353118209024331948"
    }

Plan: 3 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 3 to add, 0 to change, 0 to destroy.


10. project: git dir: tofu/git workspace: default

Plan Error

Show Output
running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu plan -input=false -refresh -out /var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git/git-default.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git': exit status 1
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 18.89s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
data.external.identity_gabrielx: Reading...
data.external.identity_arom1a: Reading...
data.external.identity_kritdass: Reading...
data.external.identity_jc_965: Reading...
data.external.identity_tentype: Reading...
data.external.identity_gostmeaper: Reading...
data.external.identity_jessica_lr_song: Reading...
data.external.identity_annadavi: Reading...
data.external.identity_jc_965: Read complete after 0s [id=-]
data.external.identity_gabrielx: Read complete after 0s [id=-]
data.external.identity_jw8: Reading...
data.external.identity_xboxbedrock: Reading...
data.external.identity_kritdass: Read complete after 0s [id=-]
data.external.identity_gostmeaper: Read complete after 0s [id=-]
data.external.identity_tentype: Read complete after 0s [id=-]
data.external.identity_jiyas: Reading...
data.external.identity_rzhu3: Reading...
data.external.identity_stringatlas: Reading...
data.external.identity_arom1a: Read complete after 0s [id=-]
data.external.identity_jessica_lr_song: Read complete after 0s [id=-]
data.external.identity_liamwalk: Reading...
data.external.identity_thesuperrl: Reading...
data.external.identity_annadavi: Read complete after 1s [id=-]
data.external.identity_anish: Reading...
data.external.identity_stringatlas: Read complete after 1s [id=-]
data.external.identity_jw8: Read complete after 1s [id=-]
data.external.identity_kluoxx: Reading...
data.external.identity_xboxbedrock: Read complete after 1s [id=-]
data.external.identity_acchs: Reading...
data.external.identity_kothurvidur: Reading...
data.external.identity_rzhu3: Read complete after 1s [id=-]
data.external.identity_liamwalk: Read complete after 1s [id=-]
data.external.identity_dcrodrig: Reading...
data.external.identity_thesuperrl: Read complete after 1s [id=-]
data.external.identity_rhsen: Reading...
data.external.identity_krishsax: Reading...
data.external.identity_jiyas: Read complete after 1s [id=-]
data.external.identity_amzoeee: Reading...
data.external.identity_manrajs: Reading...
data.external.identity_maybe_yiyi: Reading...
data.external.identity_anish: Read complete after 0s [id=-]
data.external.identity_nikhilr: Reading...
data.external.identity_kothurvidur: Read complete after 0s [id=-]
data.external.identity_acchs: Read complete after 0s [id=-]
data.external.identity_dkou: Reading...
data.external.identity_brightz: Reading...
data.external.identity_dcrodrig: Read complete after 0s [id=-]
data.external.identity_krishsax: Read complete after 0s [id=-]
data.external.identity_etashjha: Reading...
data.external.identity_kluoxx: Read complete after 0s [id=-]
data.external.identity_zoexue: Reading...
data.external.identity_amzoeee: Read complete after 0s [id=-]
data.external.identity_riztofu: Reading...
data.external.identity_emmawong: Reading...
data.external.identity_rhsen: Read complete after 0s [id=-]
data.external.identity_qianxuege: Reading...
data.external.identity_manrajs: Read complete after 0s [id=-]
data.external.identity_jalenluorion: Reading...
data.external.identity_maybe_yiyi: Read complete after 0s [id=-]
data.external.identity_kechezon: Reading...
data.external.identity_nikhilr: Read complete after 0s [id=-]
data.external.identity_siddharthr: Reading...
data.external.identity_qianxuege: Read complete after 0s [id=-]
data.external.identity_brightz: Read complete after 0s [id=-]
data.external.identity_dkou: Read complete after 0s [id=-]
random_password.kennel_webhook_secret: Refreshing state... [id=none]
forgejo_repository.cmu-insta_insta: Refreshing state... [name=insta]
forgejo_repository.devops_collie: Refreshing state... [name=collie]
forgejo_repository.uia_scottylabs_org: Refreshing state... [name=scottylabs-org]
data.external.identity_etashjha: Read complete after 0s [id=-]
data.external.identity_riztofu: Read complete after 0s [id=-]
data.external.identity_zoexue: Read complete after 0s [id=-]
data.external.identity_emmawong: Read complete after 0s [id=-]
forgejo_repository.cbp_discord_verify: Refreshing state... [name=discord-verify]
forgejo_repository.devops_documentation: Refreshing state... [name=documentation]
data.external.identity_jalenluorion: Read complete after 0s [id=-]
forgejo_repository.slai_cmugpt_agent: Refreshing state... [name=cmugpt-agent]
forgejo_repository.slai_cmugpt_sms_surface: Refreshing state... [name=cmugpt-sms-surface]
forgejo_repository.slai_cmugpt_surface: Refreshing state... [name=cmugpt-surface]
data.external.identity_kechezon: Read complete after 0s [id=-]
forgejo_repository.lost-and-found_lost_and_found_v2: Refreshing state... [name=lost-and-found-v2]
data.external.identity_siddharthr: Read complete after 1s [id=-]
forgejo_repository.cmu-housing_housing: Refreshing state... [name=housing]
forgejo_repository.cmu-eats_eats: Refreshing state... [name=eats]
forgejo_repository.cmu-eats_dining_api: Refreshing state... [name=dining-api]
forgejo_repository.slai_mcp_server: Refreshing state... [name=mcp-server]
forgejo_repository.devops_kennel: Refreshing state... [name=kennel]
forgejo_repository.cbp_bus_sign: Refreshing state... [name=bus-sign]
forgejo_repository.cmu-cal_cal: Refreshing state... [name=cal]
forgejo_repository.uia_components: Refreshing state... [name=components]
forgejo_repository.occ_occ: Refreshing state... [name=occ]
forgejo_repository.quest_quest: Refreshing state... [name=quest]
forgejo_repository.cbp_groupme_mirror: Refreshing state... [name=groupme-mirror]
data.forgejo_organization.this: Reading...
data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs]
forgejo_repository.cmu-courses_courses: Refreshing state... [name=courses]
forgejo_repository.terrier_terrier: Refreshing state... [name=terrier]
forgejo_repository.cbp_dalmatian: Refreshing state... [name=dalmatian]
forgejo_repository.devops_link_shortener: Refreshing state... [name=link-shortener]
forgejo_repository.devops_infrastructure: Refreshing state... [name=infrastructure]
forgejo_repository.cmu-maps_maps: Refreshing state... [name=maps]
forgejo_repository.cmu-courses_internet_archive: Refreshing state... [name=internet-archive]
forgejo_repository.devops_governance: Refreshing state... [name=governance]
forgejo_repository.tartan-vote_tartan_vote: Refreshing state... [name=tartan-vote]
github_repository.cmu-eats_eats: Refreshing state... [id=eats]
github_team.cmu-cal: Refreshing state... [id=15669112]
github_repository.cmu-maps_maps: Refreshing state... [id=maps]
github_repository.cmu-insta_insta: Refreshing state... [id=insta]
github_repository.slai_cmugpt_sms_surface: Refreshing state... [id=cmugpt-sms-surface]
github_team.tartan-vote: Refreshing state... [id=17701833]
github_repository.slai_mcp_server: Refreshing state... [id=mcp-server]
github_team.cmu-eats: Refreshing state... [id=16807600]
github_repository.devops_documentation: Refreshing state... [id=documentation]
github_team.occ: Refreshing state... [id=19240310]
github_repository.uia_scottylabs_org: Refreshing state... [id=scottylabs-org]
github_repository.cbp_groupme_mirror: Refreshing state... [id=groupme-mirror]
github_repository.cmu-cal_cal: Refreshing state... [id=cal]
github_repository.cbp_discord_verify: Refreshing state... [id=discord-verify]
github_repository.cmu-eats_dining_api: Refreshing state... [id=dining-api]
github_team.cmu-insta: Refreshing state... [id=19240311]
github_team.cmu-housing: Refreshing state... [id=17937919]
github_repository.devops_kennel: Refreshing state... [id=kennel]
github_repository.uia_components: Refreshing state... [id=components]
github_repository.devops_infrastructure: Refreshing state... [id=infrastructure]
github_repository.cmu-courses_internet_archive: Refreshing state... [id=internet-archive]
github_team.slai: Refreshing state... [id=17571328]
github_team.cmu-courses: Refreshing state... [id=3643907]
github_team.quest: Refreshing state... [id=17330197]
github_repository.quest_quest: Refreshing state... [id=quest]
github_repository.occ_occ: Refreshing state... [id=occ]
github_repository.devops_link_shortener: Refreshing state... [id=link-shortener]
github_repository.cmu-courses_courses: Refreshing state... [id=courses]
github_team.cmu-maps: Refreshing state... [id=10872360]
github_team.devops: Refreshing state... [id=14706218]
github_repository.slai_cmugpt_surface: Refreshing state... [id=cmugpt-surface]
github_repository.devops_governance: Refreshing state... [id=governance]
github_team.terrier: Refreshing state... [id=12028577]
github_repository.tartan-vote_tartan_vote: Refreshing state... [id=tartan-vote]
github_team.uia: Refreshing state... [id=17529763]
github_team.lost-and-found: Refreshing state... [id=4088608]
github_repository.terrier_terrier: Refreshing state... [id=terrier]
github_repository.slai_cmugpt_agent: Refreshing state... [id=cmugpt-agent]
github_repository.devops_collie: Refreshing state... [id=collie]
github_team.cbp: Refreshing state... [id=17170099]
github_repository.cbp_bus_sign: Refreshing state... [id=bus-sign]
github_repository.cmu-housing_housing: Refreshing state... [id=housing]
github_repository.cbp_dalmatian: Refreshing state... [id=dalmatian]
github_repository.lost-and-found_lost_and_found_v2: Refreshing state... [id=lost-and-found-v2]
restapi_object.devops_collie_push_mirror: Refreshing state... [id=remote_mirror_ljvVejNFXaS]
restapi_object.cmu-insta_insta_push_mirror: Refreshing state... [id=remote_mirror_891nu6XQ1us]
restapi_object.uia_scottylabs_org_push_mirror: Refreshing state... [id=remote_mirror_h6yN6QdS1lD]
restapi_object.slai_cmugpt_sms_surface_push_mirror: Refreshing state... [id=remote_mirror__fkBkASN2jA]
restapi_object.cbp_discord_verify_push_mirror: Refreshing state... [id=remote_mirror_3MV7Up6cY1X]
restapi_object.devops_documentation_push_mirror: Refreshing state... [id=remote_mirror_izmsqaNB6Wx]
restapi_object.slai_cmugpt_surface_push_mirror: Refreshing state... [id=remote_mirror_cPTQN6S5SpB]
restapi_object.lost-and-found_lost_and_found_v2_push_mirror: Refreshing state... [id=remote_mirror_n_UZBFU-CY4]
restapi_object.slai_cmugpt_agent_push_mirror: Refreshing state... [id=remote_mirror_kNY6rVcxdyr]
restapi_object.cmu-cal_cal_push_mirror: Refreshing state... [id=remote_mirror_CO243uIONoq]
forgejo_team.cmu-maps: Refreshing state... [name=cmu-maps]
forgejo_team.courses_leads: Refreshing state... [name=courses-leads]
forgejo_team.tartan-vote: Refreshing state... [name=tartan-vote]
forgejo_team.cmugpt-sms-surface_leads: Refreshing state... [name=cmugpt-sms-surface-leads]
forgejo_team.devops_leads: Refreshing state... [name=devops-leads]
forgejo_team.insta_leads: Refreshing state... [name=insta-leads]
forgejo_team.uia: Refreshing state... [name=uia]
forgejo_team.cmugpt-surface_leads: Refreshing state... [name=cmugpt-surface-leads]
forgejo_team.cmu-cal: Refreshing state... [name=cmu-cal]
forgejo_team.cbp: Refreshing state... [name=cbp]
forgejo_team.cmu-insta: Refreshing state... [name=cmu-insta]
forgejo_team.cmu-housing: Refreshing state... [name=cmu-housing]
forgejo_team.terrier: Refreshing state... [name=terrier]
forgejo_team.components_leads: Refreshing state... [name=components-leads]
forgejo_team.link-shortener_leads: Refreshing state... [name=link-shortener-leads]
forgejo_team.cmu-courses: Refreshing state... [name=cmu-courses]
forgejo_team.housing_leads: Refreshing state... [name=housing-leads]
forgejo_team.lost-and-found-v2_leads: Refreshing state... [name=lost-and-found-v2-leads]
restapi_object.cmu-housing_housing_push_mirror: Refreshing state... [id=remote_mirror_uoV2HG4aiOT]
restapi_object.cmu-housing_housing_docs_webhook: Refreshing state... [id=15]
forgejo_team.quest_leads: Refreshing state... [name=quest-leads]
forgejo_team.groupme-mirror_leads: Refreshing state... [name=groupme-mirror-leads]
forgejo_team.quest: Refreshing state... [name=quest]
forgejo_team.dalmatian_leads: Refreshing state... [name=dalmatian-leads]
forgejo_team.cmu-eats: Refreshing state... [name=cmu-eats]
forgejo_team.cmu-courses_leads: Refreshing state... [name=cmu-courses-leads]
forgejo_team.occ-main_leads: Refreshing state... [name=occ-main-leads]
forgejo_team.bus-sign_leads: Refreshing state... [name=bus-sign-leads]
forgejo_team.cmugpt-mcp-server_leads: Refreshing state... [name=cmugpt-mcp-server-leads]
forgejo_team.tartan-vote_leads: Refreshing state... [name=tartan-vote-leads]
forgejo_team.cal_leads: Refreshing state... [name=cal-leads]
forgejo_team.scottylabs-org_leads: Refreshing state... [name=scottylabs-org-leads]
forgejo_team.cmugpt-agent_leads: Refreshing state... [name=cmugpt-agent-leads]
forgejo_team.eats_leads: Refreshing state... [name=eats-leads]
forgejo_team.devops: Refreshing state... [name=devops]
forgejo_team.occ: Refreshing state... [name=occ]
forgejo_team.terrier_leads: Refreshing state... [name=terrier-leads]
forgejo_team.slai: Refreshing state... [name=slai]
forgejo_team.discord-verify_leads: Refreshing state... [name=discord-verify-leads]
forgejo_team.lost-and-found: Refreshing state... [name=lost-and-found]
forgejo_team.maps_leads: Refreshing state... [name=maps-leads]
forgejo_team.documentation_leads: Refreshing state... [name=documentation-leads]
restapi_object.slai_mcp_server_push_mirror: Refreshing state... [id=remote_mirror_Tqh2y0R_vDp]
restapi_object.devops_kennel_push_mirror: Refreshing state... [id=remote_mirror_cT9psJkhTQ4]
restapi_object.cbp_bus_sign_push_mirror: Refreshing state... [id=remote_mirror_MkKxQ2vA9pc]
restapi_object.quest_quest_push_mirror: Refreshing state... [id=remote_mirror_U9d9BPAon29]
restapi_object.uia_components_push_mirror: Refreshing state... [id=remote_mirror_Kwwa8TDnTgc]
restapi_object.occ_occ_push_mirror: Refreshing state... [id=remote_mirror_moBLq1YG1qz]
restapi_object.cbp_groupme_mirror_push_mirror: Refreshing state... [id=remote_mirror_U61HJHB9VGE]
restapi_object.cmu-courses_courses_push_mirror: Refreshing state... [id=remote_mirror_WMR9yQwKFDF]
restapi_object.terrier_terrier_push_mirror: Refreshing state... [id=remote_mirror_YopDgZksZv5]
restapi_object.devops_link_shortener_push_mirror: Refreshing state... [id=remote_mirror_cp8daMpt82v]
restapi_object.cbp_dalmatian_push_mirror: Refreshing state... [id=remote_mirror_qzlCpLK153F]
restapi_object.cbp_dalmatian_docs_webhook: Refreshing state... [id=7]
restapi_object.devops_infrastructure_push_mirror: Refreshing state... [id=remote_mirror_oYtFpKVzTaZ]
restapi_object.cmu-courses_internet_archive_push_mirror: Refreshing state... [id=remote_mirror_58X__Yfs0bR]
restapi_object.devops_governance_push_mirror: Refreshing state... [id=remote_mirror_lt9JzkX1jMq]
restapi_object.cmu-maps_maps_push_mirror: Refreshing state... [id=remote_mirror_e4_FUnvoslJ]
restapi_object.tartan-vote_tartan_vote_push_mirror: Refreshing state... [id=remote_mirror_EUFq4sFSnH5]
restapi_object.tartan-vote_tartan_vote_docs_webhook: Refreshing state... [id=25]
github_team_membership.cmu-cal_anish: Refreshing state... [id=15669112:ap-1]
github_team_membership.cmu-cal_jalenluorion: Refreshing state... [id=15669112:jalenluorion]
github_actions_repository_permissions.slai_cmugpt_sms_surface_actions: Refreshing state... [id=cmugpt-sms-surface]
github_team_membership.tartan-vote_xboxbedrock: Refreshing state... [id=17701833:xboxbedrock]
github_team_membership.tartan-vote_anish: Refreshing state... [id=17701833:ap-1]
github_team_membership.tartan-vote_riztofu: Refreshing state... [id=17701833:riztofu]
github_team_membership.tartan-vote_tentype: Refreshing state... [id=17701833:tentype]
github_team_membership.tartan-vote_maybe_yiyi: Refreshing state... [id=17701833:maybe-yiyi]
github_team_membership.cmu-eats_arom1a: Refreshing state... [id=16807600:arom1a]
github_team_membership.occ_xboxbedrock: Refreshing state... [id=19240310:xboxbedrock]
github_team_membership.occ_dcrodrig: Refreshing state... [id=19240310:newhacker1746]
github_team_membership.occ_anish: Refreshing state... [id=19240310:ap-1]
github_team_membership.occ_acchs: Refreshing state... [id=19240310:acchs-at-cmu]
github_actions_repository_permissions.cmu-eats_eats_actions: Refreshing state... [id=eats]
github_actions_repository_permissions.cmu-maps_maps_actions: Refreshing state... [id=maps]
github_actions_repository_permissions.cmu-insta_insta_actions: Refreshing state... [id=insta]
github_actions_repository_permissions.cbp_groupme_mirror_actions: Refreshing state... [id=groupme-mirror]
github_actions_repository_permissions.slai_mcp_server_actions: Refreshing state... [id=mcp-server]
github_actions_repository_permissions.cbp_discord_verify_actions: Refreshing state... [id=discord-verify]
github_actions_repository_permissions.devops_documentation_actions: Refreshing state... [id=documentation]
github_team_membership.cmu-insta_etashjha: Refreshing state... [id=19240311:etashj]
github_actions_repository_permissions.uia_scottylabs_org_actions: Refreshing state... [id=scottylabs-org]
github_team_membership.cmu-housing_emmawong: Refreshing state... [id=17937919:emw868]
github_team_membership.cmu-housing_anish: Refreshing state... [id=17937919:ap-1]
github_team_membership.cmu-housing_nikhilr: Refreshing state... [id=17937919:ecstaticpilot]
github_team_membership.cmu-housing_tentype: Refreshing state... [id=17937919:tentype]
github_team_membership.cmu-housing_zoexue: Refreshing state... [id=17937919:o-zox-o]
github_team_membership.cmu-housing_xboxbedrock: Refreshing state... [id=17937919:xboxbedrock]
github_team_membership.cmu-housing_gostmeaper: Refreshing state... [id=17937919:gostmeaper]
github_team_membership.cmu-housing_jessica_lr_song: Refreshing state... [id=17937919:jessica-lr-song]
github_team_membership.cmu-housing_manrajs: Refreshing state... [id=17937919:manraj10]
github_team_membership.cmu-housing_amzoeee: Refreshing state... [id=17937919:amzoeee]
github_team_membership.cmu-housing_kluoxx: Refreshing state... [id=17937919:luokaden888]
github_team_membership.cmu-housing_rhsen: Refreshing state... [id=17937919:rohansen08]
github_team_membership.cmu-housing_stringatlas: Refreshing state... [id=17937919:stringatlas]
github_actions_repository_permissions.uia_components_actions: Refreshing state... [id=components]
github_actions_repository_permissions.cmu-cal_cal_actions: Refreshing state... [id=cal]
github_actions_repository_permissions.cmu-courses_internet_archive_actions: Refreshing state... [id=internet-archive]
github_team_membership.slai_jc_965: Refreshing state... [id=17571328:jc-965]
github_team_membership.slai_anish: Refreshing state... [id=17571328:ap-1]
github_team_membership.slai_jiyas: Refreshing state... [id=17571328:jiya2301]
github_team_membership.slai_xboxbedrock: Refreshing state... [id=17571328:xboxbedrock]
github_team_membership.slai_qianxuege: Refreshing state... [id=17571328:qianxuege]
github_team_membership.slai_krishsax: Refreshing state... [id=17571328:krishsaxena]
github_team_membership.slai_rzhu3: Refreshing state... [id=17571328:richard-31415]
github_actions_repository_permissions.cmu-eats_dining_api_actions: Refreshing state... [id=dining-api]
github_team_membership.cmu-courses_anish: Refreshing state... [id=3643907:ap-1]
github_team_membership.cmu-courses_etashjha: Refreshing state... [id=3643907:etashj]
github_team_membership.cmu-courses_dkou: Refreshing state... [id=3643907:sharkree]
github_team_membership.cmu-courses_gabrielx: Refreshing state... [id=3643907:gabriel-xu]
github_team_membership.quest_xboxbedrock: Refreshing state... [id=17330197:xboxbedrock]
github_team_membership.quest_brightz: Refreshing state... [id=17330197:bright-zheng]
github_team_membership.quest_jw8: Refreshing state... [id=17330197:jeffreyyijiwang]
github_team_membership.quest_kechezon: Refreshing state... [id=17330197:kechezon]
github_team_membership.quest_anish: Refreshing state... [id=17330197:ap-1]
github_actions_repository_permissions.devops_kennel_actions: Refreshing state... [id=kennel]
github_actions_repository_permissions.devops_infrastructure_actions: Refreshing state... [id=infrastructure]
github_actions_repository_permissions.cmu-courses_courses_actions: Refreshing state... [id=courses]
github_team_membership.cmu-maps_anish: Refreshing state... [id=10872360:ap-1]
github_team_membership.cmu-maps_siddharthr: Refreshing state... [id=10872360:overthrowing]
github_team_membership.devops_anish: Refreshing state... [id=14706218:ap-1]
github_team_membership.devops_thesuperrl: Refreshing state... [id=14706218:thesuperrl]
github_team_membership.devops_xboxbedrock: Refreshing state... [id=14706218:xboxbedrock]
github_actions_repository_permissions.slai_cmugpt_surface_actions: Refreshing state... [id=cmugpt-surface]
github_actions_repository_permissions.quest_quest_actions: Refreshing state... [id=quest]
github_team_membership.terrier_kothurvidur: Refreshing state... [id=12028577:kothurvidur]
github_team_membership.terrier_dkou: Refreshing state... [id=12028577:sharkree]
github_team_membership.terrier_kritdass: Refreshing state... [id=12028577:kritdass]
github_team_membership.terrier_anish: Refreshing state... [id=12028577:ap-1]
github_team_membership.terrier_amzoeee: Refreshing state... [id=12028577:amzoeee]
github_team_membership.terrier_etashjha: Refreshing state... [id=12028577:etashj]
github_actions_repository_permissions.occ_occ_actions: Refreshing state... [id=occ]
github_actions_repository_permissions.devops_link_shortener_actions: Refreshing state... [id=link-shortener]
github_team_membership.uia_anish: Refreshing state... [id=17529763:ap-1]
github_team_membership.lost-and-found_anish: Refreshing state... [id=4088608:ap-1]
github_actions_repository_permissions.slai_cmugpt_agent_actions: Refreshing state... [id=cmugpt-agent]
restapi_object.slai_mcp_server_kennel_webhook: Refreshing state... [id=13]
restapi_object.uia_components_kennel_webhook: Refreshing state... [id=27]
restapi_object.devops_collie_kennel_webhook: Refreshing state... [id=17]
restapi_object.slai_cmugpt_sms_surface_kennel_webhook: Refreshing state... [id=19]
restapi_object.tartan-vote_tartan_vote_kennel_webhook: Refreshing state... [id=24]
restapi_object.cbp_discord_verify_kennel_webhook: Refreshing state... [id=18]
restapi_object.quest_quest_kennel_webhook: Refreshing state... [id=10]
restapi_object.devops_kennel_kennel_webhook: Refreshing state... [id=14]
restapi_object.cbp_bus_sign_kennel_webhook: Refreshing state... [id=21]
restapi_object.cmu-maps_maps_kennel_webhook: Refreshing state... [id=9]
restapi_object.cbp_dalmatian_kennel_webhook: Refreshing state... [id=6]
restapi_object.uia_scottylabs_org_kennel_webhook: Refreshing state... [id=29]
restapi_object.devops_governance_kennel_webhook: Refreshing state... [id=22]
restapi_object.cmu-cal_cal_kennel_webhook: Refreshing state... [id=20]
restapi_object.cmu-courses_courses_kennel_webhook: Refreshing state... [id=11]
restapi_object.slai_cmugpt_surface_kennel_webhook: Refreshing state... [id=23]
restapi_object.cmu-housing_housing_kennel_webhook: Refreshing state... [id=16]
restapi_object.terrier_terrier_kennel_webhook: Refreshing state... [id=12]
restapi_object.slai_cmugpt_agent_kennel_webhook: Refreshing state... [id=28]
restapi_object.devops_link_shortener_kennel_webhook: Refreshing state... [id=8]
restapi_object.devops_infrastructure_kennel_webhook: Refreshing state... [id=26]
github_repository_deploy_key.cmu-insta_insta_mirror_deploy_key: Refreshing state... [id=insta:161720245]
github_repository_deploy_key.slai_cmugpt_sms_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-sms-surface:159665365]
github_repository_deploy_key.uia_scottylabs_org_mirror_deploy_key: Refreshing state... [id=scottylabs-org:159665361]
github_repository_deploy_key.cbp_discord_verify_mirror_deploy_key: Refreshing state... [id=discord-verify:159665340]
github_actions_repository_permissions.devops_governance_actions: Refreshing state... [id=governance]
github_repository_deploy_key.devops_documentation_mirror_deploy_key: Refreshing state... [id=documentation:159665362]
github_repository_deploy_key.slai_cmugpt_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-surface:159665348]
github_repository_deploy_key.slai_cmugpt_agent_mirror_deploy_key: Refreshing state... [id=cmugpt-agent:159665360]
github_repository_deploy_key.cmu-cal_cal_mirror_deploy_key: Refreshing state... [id=cal:159665344]
forgejo_team_member.cmu-maps_siddharthr: Refreshing state...
forgejo_team_member.cmu-maps_anish: Refreshing state...
forgejo_team_repository.cmu-maps_maps: Refreshing state...
forgejo_team_member.tartan-vote_anish: Refreshing state...
forgejo_team_member.tartan-vote_maybe_yiyi: Refreshing state...
forgejo_team_repository.tartan-vote_tartan_vote: Refreshing state...
forgejo_team_member.tartan-vote_xboxbedrock: Refreshing state...
forgejo_team_member.courses_leads_anish: Refreshing state...
forgejo_team_member.courses_leads_etashjha: Refreshing state...
forgejo_team_member.tartan-vote_tentype: Refreshing state...
forgejo_team_member.tartan-vote_riztofu: Refreshing state...
forgejo_team_member.devops_leads_thesuperrl: Refreshing state...
forgejo_team_member.devops_leads_anish: Refreshing state...
forgejo_team_member.cmugpt-sms-surface_leads_krishsax: Refreshing state...
forgejo_team_member.cmugpt-sms-surface_leads_anish: Refreshing state...
forgejo_team_member.insta_leads_etashjha: Refreshing state...
forgejo_team_member.uia_anish: Refreshing state...
forgejo_team_repository.uia_scottylabs_org: Refreshing state...
forgejo_team_repository.uia_components: Refreshing state...
forgejo_team_member.cmugpt-surface_leads_anish: Refreshing state...
forgejo_team_member.cmugpt-surface_leads_krishsax: Refreshing state...
forgejo_team_member.cmu-cal_anish: Refreshing state...
forgejo_team_member.cmu-cal_jalenluorion: Refreshing state...
forgejo_team_repository.cmu-cal_cal: Refreshing state...
forgejo_team_member.cmu-insta_etashjha: Refreshing state...
forgejo_team_repository.cmu-insta_insta: Refreshing state...
forgejo_team_member.cbp_tentype: Refreshing state...
forgejo_team_member.cbp_xboxbedrock: Refreshing state...
forgejo_team_member.cbp_anish: Refreshing state...
forgejo_team_member.cbp_liamwalk: Refreshing state...
forgejo_team_repository.cbp_discord_verify: Refreshing state...
forgejo_team_member.cbp_maybe_yiyi: Refreshing state...
forgejo_team_repository.cbp_dalmatian: Refreshing state...
forgejo_team_member.cbp_nikhilr: Refreshing state...
forgejo_team_repository.cbp_bus_sign: Refreshing state...
forgejo_team_repository.cbp_groupme_mirror: Refreshing state...
forgejo_team_member.cmu-housing_tentype: Refreshing state...
forgejo_team_member.terrier_amzoeee: Refreshing state...
forgejo_team_member.cmu-housing_zoexue: Refreshing state...
forgejo_team_member.terrier_anish: Refreshing state...
forgejo_team_member.cmu-housing_kluoxx: Refreshing state...
forgejo_team_member.cmu-housing_anish: Refreshing state...
forgejo_team_member.cmu-housing_stringatlas: Refreshing state...
forgejo_team_member.cmu-housing_rhsen: Refreshing state...
forgejo_team_member.terrier_kothurvidur: Refreshing state...
forgejo_team_member.terrier_kritdass: Refreshing state...
forgejo_team_repository.terrier_terrier: Refreshing state...
forgejo_team_member.cmu-housing_xboxbedrock: Refreshing state...
forgejo_team_member.terrier_etashjha: Refreshing state...
forgejo_team_member.cmu-housing_nikhilr: Refreshing state...
forgejo_team_member.cmu-housing_emmawong: Refreshing state...
forgejo_team_repository.cmu-housing_housing: Refreshing state...
forgejo_team_member.cmu-housing_jessica_lr_song: Refreshing state...
forgejo_team_member.cmu-housing_amzoeee: Refreshing state...
forgejo_team_member.cmu-housing_manrajs: Refreshing state...
forgejo_team_member.cmu-housing_gostmeaper: Refreshing state...
forgejo_team_member.terrier_dkou: Refreshing state...
forgejo_team_member.link-shortener_leads_thesuperrl: Refreshing state...
forgejo_team_member.link-shortener_leads_anish: Refreshing state...
forgejo_team_repository.cmu-courses_internet_archive: Refreshing state...
forgejo_team_repository.cmu-courses_courses: Refreshing state...
forgejo_team_member.cmu-courses_etashjha: Refreshing state...
forgejo_team_member.cmu-courses_dkou: Refreshing state...
forgejo_team_member.cmu-courses_anish: Refreshing state...
forgejo_team_member.cmu-courses_gabrielx: Refreshing state...
forgejo_team_member.components_leads_anish: Refreshing state...
forgejo_team_member.housing_leads_tentype: Refreshing state...
forgejo_team_member.housing_leads_gostmeaper: Refreshing state...
forgejo_team_member.housing_leads_emmawong: Refreshing state...
forgejo_team_member.housing_leads_anish: Refreshing state...
forgejo_team_member.housing_leads_nikhilr: Refreshing state...
forgejo_team_member.lost-and-found-v2_leads_anish: Refreshing state...
github_actions_repository_permissions.devops_collie_actions: Refreshing state... [id=collie]
github_repository_deploy_key.devops_collie_mirror_deploy_key: Refreshing state... [id=collie:159665332]
forgejo_team_member.quest_leads_xboxbedrock: Refreshing state...
forgejo_team_member.quest_leads_anish: Refreshing state...
forgejo_team_member.groupme-mirror_leads_anish: Refreshing state...
forgejo_team_member.groupme-mirror_leads_maybe_yiyi: Refreshing state...
forgejo_team_repository.quest_quest: Refreshing state...
forgejo_team_member.groupme-mirror_leads_tentype: Refreshing state...
forgejo_team_member.quest_kechezon: Refreshing state...
forgejo_team_member.quest_jw8: Refreshing state...
forgejo_team_member.quest_brightz: Refreshing state...
forgejo_team_member.quest_anish: Refreshing state...
forgejo_team_member.quest_xboxbedrock: Refreshing state...
forgejo_team_member.dalmatian_leads_tentype: Refreshing state...
forgejo_team_member.dalmatian_leads_anish: Refreshing state...
forgejo_team_member.dalmatian_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.cmu-eats_arom1a: Refreshing state...
forgejo_team_member.cmu-courses_leads_etashjha: Refreshing state...
forgejo_team_member.cmu-courses_leads_anish: Refreshing state...
forgejo_team_member.occ-main_leads_dcrodrig: Refreshing state...
forgejo_team_member.occ-main_leads_anish: Refreshing state...
forgejo_team_member.bus-sign_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.bus-sign_leads_tentype: Refreshing state...
forgejo_team_member.bus-sign_leads_anish: Refreshing state...
forgejo_team_member.cmugpt-mcp-server_leads_anish: Refreshing state...
forgejo_team_member.cmugpt-mcp-server_leads_krishsax: Refreshing state...
forgejo_team_member.tartan-vote_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.tartan-vote_leads_anish: Refreshing state...
forgejo_team_member.scottylabs-org_leads_anish: Refreshing state...
forgejo_team_member.cal_leads_jalenluorion: Refreshing state...
forgejo_team_member.cal_leads_anish: Refreshing state...
forgejo_team_member.cmugpt-agent_leads_anish: Refreshing state...
forgejo_team_member.cmugpt-agent_leads_krishsax: Refreshing state...
github_team_membership.cbp_liamwalk: Refreshing state... [id=17170099:pandaaured]
github_team_membership.cbp_nikhilr: Refreshing state... [id=17170099:ecstaticpilot]
github_team_membership.cbp_anish: Refreshing state... [id=17170099:ap-1]
github_team_membership.cbp_tentype: Refreshing state... [id=17170099:tentype]
github_team_membership.cbp_maybe_yiyi: Refreshing state... [id=17170099:maybe-yiyi]
github_team_membership.cbp_xboxbedrock: Refreshing state... [id=17170099:xboxbedrock]
forgejo_team_repository.devops_link_shortener: Refreshing state...
forgejo_team_repository.devops_kennel: Refreshing state...
forgejo_team_repository.devops_governance: Refreshing state...
forgejo_team_repository.devops_infrastructure: Refreshing state...
forgejo_team_member.devops_thesuperrl: Refreshing state...
forgejo_team_member.devops_anish: Refreshing state...
forgejo_team_member.devops_xboxbedrock: Refreshing state...
forgejo_team_member.eats_leads_arom1a: Refreshing state...
forgejo_team_repository.devops_documentation: Refreshing state...
forgejo_team_repository.devops_collie: Refreshing state...
forgejo_team_member.slai_anish: Refreshing state...
forgejo_team_member.slai_jiyas: Refreshing state...
forgejo_team_repository.slai_cmugpt_sms_surface: Refreshing state...
forgejo_team_repository.slai_mcp_server: Refreshing state...
forgejo_team_member.slai_jc_965: Refreshing state...
forgejo_team_member.slai_krishsax: Refreshing state...
forgejo_team_member.slai_qianxuege: Refreshing state...
forgejo_team_member.slai_rzhu3: Refreshing state...
forgejo_team_member.slai_xboxbedrock: Refreshing state...
forgejo_team_repository.slai_cmugpt_surface: Refreshing state...
forgejo_team_member.occ_xboxbedrock: Refreshing state...
forgejo_team_member.occ_dcrodrig: Refreshing state...
forgejo_team_repository.occ_occ: Refreshing state...
forgejo_team_member.occ_acchs: Refreshing state...
forgejo_team_member.occ_anish: Refreshing state...
forgejo_team_member.terrier_leads_kritdass: Refreshing state...
forgejo_team_member.terrier_leads_anish: Refreshing state...
forgejo_team_repository.slai_cmugpt_agent: Refreshing state...
forgejo_team_member.discord-verify_leads_tentype: Refreshing state...
forgejo_team_member.discord-verify_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.discord-verify_leads_anish: Refreshing state...
forgejo_team_repository.lost-and-found_lost_and_found_v2: Refreshing state...
forgejo_team_member.lost-and-found_anish: Refreshing state...
forgejo_team_member.maps_leads_siddharthr: Refreshing state...
forgejo_team_member.maps_leads_anish: Refreshing state...
forgejo_team_member.documentation_leads_thesuperrl: Refreshing state...
forgejo_team_member.documentation_leads_anish: Refreshing state...
github_repository_deploy_key.devops_kennel_mirror_deploy_key: Refreshing state... [id=kennel:159665328]
github_repository_deploy_key.slai_mcp_server_mirror_deploy_key: Refreshing state... [id=mcp-server:159665358]
github_repository_deploy_key.quest_quest_mirror_deploy_key: Refreshing state... [id=quest:159665323]
github_repository_deploy_key.occ_occ_mirror_deploy_key: Refreshing state... [id=occ:161720242]
github_repository_deploy_key.uia_components_mirror_deploy_key: Refreshing state... [id=components:159665353]
github_repository_deploy_key.cbp_groupme_mirror_mirror_deploy_key: Refreshing state... [id=groupme-mirror:159665366]
github_repository_deploy_key.cmu-courses_courses_mirror_deploy_key: Refreshing state... [id=courses:159665327]
github_repository_deploy_key.devops_link_shortener_mirror_deploy_key: Refreshing state... [id=link-shortener:159665345]
github_repository_deploy_key.devops_infrastructure_mirror_deploy_key: Refreshing state... [id=infrastructure:159665343]
github_repository_deploy_key.cmu-courses_internet_archive_mirror_deploy_key: Refreshing state... [id=internet-archive:159665356]
github_repository_deploy_key.devops_governance_mirror_deploy_key: Refreshing state... [id=governance:159665346]
github_repository_deploy_key.cmu-maps_maps_mirror_deploy_key: Refreshing state... [id=maps:159665325]
github_actions_repository_permissions.tartan-vote_tartan_vote_actions: Refreshing state... [id=tartan-vote]
github_repository_deploy_key.tartan-vote_tartan_vote_mirror_deploy_key: Refreshing state... [id=tartan-vote:159665350]
github_actions_repository_permissions.lost-and-found_lost_and_found_v2_actions: Refreshing state... [id=lost-and-found-v2]
github_repository_deploy_key.lost-and-found_lost_and_found_v2_mirror_deploy_key: Refreshing state... [id=lost-and-found-v2:159665337]
github_repository_deploy_key.terrier_terrier_mirror_deploy_key: Refreshing state... [id=terrier:159665339]
github_actions_repository_permissions.terrier_terrier_actions: Refreshing state... [id=terrier]
github_repository_deploy_key.cbp_bus_sign_mirror_deploy_key: Refreshing state... [id=bus-sign:159665342]
github_actions_repository_permissions.cbp_bus_sign_actions: Refreshing state... [id=bus-sign]
github_actions_repository_permissions.cmu-housing_housing_actions: Refreshing state... [id=housing]
github_repository_deploy_key.cmu-housing_housing_mirror_deploy_key: Refreshing state... [id=housing:159665369]
github_repository_deploy_key.cbp_dalmatian_mirror_deploy_key: Refreshing state... [id=dalmatian:159665367]
github_actions_repository_permissions.cbp_dalmatian_actions: Refreshing state... [id=dalmatian]
forgejo_team_repository.maps_leads_maps: Refreshing state...
forgejo_team_repository.tartan-vote_leads_tartan_vote: Refreshing state...
github_repository_ruleset.cmu-insta_insta_ruleset: Refreshing state... [id=21841636]
forgejo_team_repository.scottylabs-org_leads_scottylabs_org: Refreshing state...
forgejo_team_repository.components_leads_components: Refreshing state...
forgejo_team_repository.cal_leads_cal: Refreshing state...
github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Refreshing state... [id=17653622]
forgejo_team_repository.insta_leads_insta: Refreshing state...
forgejo_team_repository.discord-verify_leads_discord_verify: Refreshing state...
forgejo_team_repository.dalmatian_leads_dalmatian: Refreshing state...
forgejo_team_repository.bus-sign_leads_bus_sign: Refreshing state...
forgejo_team_repository.groupme-mirror_leads_groupme_mirror: Refreshing state...
github_repository_ruleset.uia_scottylabs_org_ruleset: Refreshing state... [id=18112214]
forgejo_team_repository.terrier_leads_terrier: Refreshing state...
forgejo_team_repository.housing_leads_housing: Refreshing state...
forgejo_team_repository.courses_leads_courses: Refreshing state...
forgejo_team_repository.cmu-courses_leads_internet_archive: Refreshing state...
github_repository_ruleset.cbp_discord_verify_ruleset: Refreshing state... [id=11449154]
forgejo_team_repository.quest_leads_quest: Refreshing state...
github_repository_ruleset.devops_documentation_ruleset: Refreshing state... [id=17079368]
github_repository_ruleset.slai_cmugpt_surface_ruleset: Refreshing state... [id=16476662]
github_repository_ruleset.slai_cmugpt_agent_ruleset: Refreshing state... [id=16476661]
forgejo_team_repository.link-shortener_leads_link_shortener: Refreshing state...
forgejo_team_repository.devops_leads_kennel: Refreshing state...
forgejo_team_repository.devops_leads_governance: Refreshing state...
forgejo_team_repository.devops_leads_infrastructure: Refreshing state...
github_repository_ruleset.cmu-cal_cal_ruleset: Refreshing state... [id=18919281]
forgejo_team_repository.documentation_leads_documentation: Refreshing state...
forgejo_team_repository.devops_leads_collie: Refreshing state...
forgejo_team_repository.cmugpt-sms-surface_leads_cmugpt_sms_surface: Refreshing state...
forgejo_team_repository.cmugpt-mcp-server_leads_mcp_server: Refreshing state...
forgejo_team_repository.cmugpt-surface_leads_cmugpt_surface: Refreshing state...
forgejo_team_repository.occ-main_leads_occ: Refreshing state...
forgejo_team_repository.cmugpt-agent_leads_cmugpt_agent: Refreshing state...
forgejo_team_repository.lost-and-found-v2_leads_lost_and_found_v2: Refreshing state...
github_repository_ruleset.devops_collie_ruleset: Refreshing state... [id=18865204]
github_repository_ruleset.devops_kennel_ruleset: Refreshing state... [id=13193453]
github_repository_ruleset.slai_mcp_server_ruleset: Refreshing state... [id=17082154]
github_repository_ruleset.quest_quest_ruleset: Refreshing state... [id=15750052]
github_repository_ruleset.occ_occ_ruleset: Refreshing state... [id=21841634]
github_repository_ruleset.uia_components_ruleset: Refreshing state... [id=16366085]
github_repository_ruleset.cbp_groupme_mirror_ruleset: Refreshing state... [id=12503296]
github_repository_ruleset.cmu-courses_courses_ruleset: Refreshing state... [id=15855161]
github_repository_ruleset.devops_link_shortener_ruleset: Refreshing state... [id=17911647]
github_repository_ruleset.devops_infrastructure_ruleset: Refreshing state... [id=11275118]
github_repository_ruleset.cmu-courses_internet_archive_ruleset: Refreshing state... [id=11988074]
github_repository_ruleset.devops_governance_ruleset: Refreshing state... [id=14488792]
github_repository_ruleset.cmu-maps_maps_ruleset: Refreshing state... [id=18864961]
github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Refreshing state... [id=16848966]
github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Refreshing state... [id=18920618]
github_repository_ruleset.terrier_terrier_ruleset: Refreshing state... [id=12671003]
github_repository_ruleset.cbp_bus_sign_ruleset: Refreshing state... [id=15236550]
github_repository_ruleset.cmu-housing_housing_ruleset: Refreshing state... [id=17421604]
github_repository_ruleset.cbp_dalmatian_ruleset: Refreshing state... [id=17238915]

OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
  + create

OpenTofu planned the following actions, but then encountered a problem:

  # forgejo_repository.cmu-study_study will be created
  + resource "forgejo_repository" "cmu-study_study" {
      + allow_manual_merge          = false
      + allow_merge_commits         = true
      + allow_rebase                = true
      + allow_rebase_explicit       = true
      + allow_squash_merge          = true
      + archive_on_destroy          = false
      + archived                    = false
      + auto_init                   = true
      + autodetect_manual_merge     = false
      + avatar_url                  = (known after apply)
      + clone_url                   = (known after apply)
      + created_at                  = (known after apply)
      + default_branch              = "main"
      + default_merge_style         = "merge"
      + description                 = "A platform for creating and joining study groups"
      + empty                       = (known after apply)
      + external_tracker            = (known after apply)
      + external_wiki               = (known after apply)
      + fork                        = (known after apply)
      + forks_count                 = (known after apply)
      + full_name                   = (known after apply)
      + has_actions                 = true
      + has_issues                  = true
      + has_packages                = true
      + has_projects                = true
      + has_pull_requests           = true
      + has_releases                = true
      + has_wiki                    = true
      + html_url                    = (known after apply)
      + id                          = (known after apply)
      + ignore_whitespace_conflicts = false
      + internal                    = (known after apply)
      + internal_tracker            = (known after apply)
      + labels                      = false
      + lfs                         = false
      + milestones                  = false
      + mirror                      = false
      + mirror_interval             = (known after apply)
      + mirror_updated              = (known after apply)
      + name                        = "study"
      + open_issues_count           = (known after apply)
      + open_pr_counter             = (known after apply)
      + owner                       = "ScottyLabs"
      + parent_id                   = (known after apply)
      + permissions                 = (known after apply)
      + private                     = false
      + release_counter             = (known after apply)
      + size                        = (known after apply)
      + ssh_url                     = (known after apply)
      + stars_count                 = (known after apply)
      + template                    = false
      + trust_model                 = "default"
      + updated_at                  = (known after apply)
      + watchers_count              = (known after apply)
      + website                     = "https://cmustudy.com"
    }

  # forgejo_team.cmu-study will be created
  + resource "forgejo_team" "cmu-study" {
      + can_create_org_repo       = (known after apply)
      + description               = "CMU Study"
      + id                        = (known after apply)
      + includes_all_repositories = (known after apply)
      + name                      = "cmu-study"
      + organization              = (known after apply)
      + organization_id           = 3
      + permission                = "write"
      + units_map                 = {
          + "repo.code"   = "write"
          + "repo.issues" = "write"
          + "repo.pulls"  = "write"
        }
    }

  # forgejo_team.study_leads will be created
  + resource "forgejo_team" "study_leads" {
      + can_create_org_repo       = false
      + description               = "CMU Study leads"
      + id                        = (known after apply)
      + includes_all_repositories = false
      + name                      = "study-leads"
      + organization              = (known after apply)
      + organization_id           = 3
      + permission                = "admin"
      + units_map                 = {
          + "repo.actions"    = "admin"
          + "repo.code"       = "admin"
          + "repo.ext_issues" = "admin"
          + "repo.ext_wiki"   = "admin"
          + "repo.issues"     = "admin"
          + "repo.packages"   = "admin"
          + "repo.projects"   = "admin"
          + "repo.pulls"      = "admin"
          + "repo.releases"   = "admin"
          + "repo.wiki"       = "admin"
        }
    }

  # forgejo_team_member.cmu-study_annadavi will be created
  + resource "forgejo_team_member" "cmu-study_annadavi" {
      + team_id = (known after apply)
      + user    = "annadavi"
    }

  # forgejo_team_member.study_leads_annadavi will be created
  + resource "forgejo_team_member" "study_leads_annadavi" {
      + team_id = (known after apply)
      + user    = "annadavi"
    }

  # forgejo_team_repository.cmu-study_study will be created
  + resource "forgejo_team_repository" "cmu-study_study" {
      + owner      = "ScottyLabs"
      + repository = "study"
      + team_id    = (known after apply)
    }

  # forgejo_team_repository.study_leads_study will be created
  + resource "forgejo_team_repository" "study_leads_study" {
      + owner      = "ScottyLabs"
      + repository = "study"
      + team_id    = (known after apply)
    }

  # github_actions_repository_permissions.cmu-study_study_actions will be created
  + resource "github_actions_repository_permissions" "cmu-study_study_actions" {
      + enabled              = false
      + id                   = (known after apply)
      + repository           = "study"
      + sha_pinning_required = (known after apply)
    }

  # github_repository.cmu-study_study will be created
  + resource "github_repository" "cmu-study_study" {
      + allow_auto_merge                        = false
      + allow_forking                           = (known after apply)
      + allow_merge_commit                      = true
      + allow_rebase_merge                      = true
      + allow_squash_merge                      = true
      + archived                                = false
      + default_branch                          = (known after apply)
      + delete_branch_on_merge                  = false
      + description                             = "READ-ONLY MIRROR"
      + etag                                    = (known after apply)
      + fork                                    = (known after apply)
      + full_name                               = (known after apply)
      + git_clone_url                           = (known after apply)
      + has_discussions                         = false
      + has_issues                              = false
      + has_projects                            = false
      + has_wiki                                = false
      + homepage_url                            = "https://git.cmu.dev/ScottyLabs/study"
      + html_url                                = (known after apply)
      + http_clone_url                          = (known after apply)
      + id                                      = (known after apply)
      + ignore_vulnerability_alerts_during_read = false
      + merge_commit_message                    = "PR_TITLE"
      + merge_commit_title                      = "MERGE_MESSAGE"
      + name                                    = "study"
      + node_id                                 = (known after apply)
      + primary_language                        = (known after apply)
      + private                                 = (known after apply)
      + repo_id                                 = (known after apply)
      + source_owner                            = (known after apply)
      + source_repo                             = (known after apply)
      + squash_merge_commit_message             = "COMMIT_MESSAGES"
      + squash_merge_commit_title               = "COMMIT_OR_PR_TITLE"
      + ssh_clone_url                           = (known after apply)
      + svn_url                                 = (known after apply)
      + topics                                  = (known after apply)
      + visibility                              = "public"
      + vulnerability_alerts                    = (known after apply)
      + web_commit_signoff_required             = (known after apply)

      + security_and_analysis (known after apply)
    }

  # github_repository_deploy_key.cmu-study_study_mirror_deploy_key will be created
  + resource "github_repository_deploy_key" "cmu-study_study_mirror_deploy_key" {
      + etag       = (known after apply)
      + id         = (known after apply)
      + key        = (known after apply)
      + read_only  = false
      + repository = "study"
      + title      = "Forgejo mirror"
    }

  # github_repository_ruleset.cmu-study_study_ruleset will be created
  + resource "github_repository_ruleset" "cmu-study_study_ruleset" {
      + enforcement = "active"
      + etag        = (known after apply)
      + id          = (known after apply)
      + name        = "Default"
      + node_id     = (known after apply)
      + repository  = "study"
      + ruleset_id  = (known after apply)
      + target      = "branch"

      + bypass_actors {
          + actor_id    = (known after apply)
          + actor_type  = "DeployKey"
          + bypass_mode = "always"
        }

      + conditions {
          + ref_name {
              + exclude = []
              + include = [
                  + "~ALL",
                ]
            }
        }

      + rules {
          + creation                      = true
          + deletion                      = true
          + non_fast_forward              = true
          + update                        = true
          + update_allows_fetch_and_merge = false

          + pull_request {
              + allowed_merge_methods             = (known after apply)
              + dismiss_stale_reviews_on_push     = false
              + require_code_owner_review         = false
              + require_last_push_approval        = false
              + required_approving_review_count   = 1
              + required_review_thread_resolution = false
            }
        }
    }

  # github_team.cmu-study will be created
  + resource "github_team" "cmu-study" {
      + create_default_maintainer = false
      + etag                      = (known after apply)
      + id                        = (known after apply)
      + members_count             = (known after apply)
      + name                      = "CMU Study"
      + node_id                   = (known after apply)
      + notification_setting      = "notifications_enabled"
      + parent_team_read_id       = (known after apply)
      + parent_team_read_slug     = (known after apply)
      + privacy                   = "closed"
      + slug                      = (known after apply)
    }

  # github_team_membership.cmu-study_annadavi will be created
  + resource "github_team_membership" "cmu-study_annadavi" {
      + etag     = (known after apply)
      + id       = (known after apply)
      + role     = "maintainer"
      + team_id  = (known after apply)
      + username = "annadavid1"
    }

  # restapi_object.cmu-study_study_kennel_webhook will be created
  + resource "restapi_object" "cmu-study_study_kennel_webhook" {
      + api_data        = (known after apply)
      + api_response    = (known after apply)
      + create_response = (known after apply)
      + data            = (sensitive value)
      + id              = (known after apply)
      + id_attribute    = "id"
      + path            = "/api/v1/repos/ScottyLabs/study/hooks"
      + update_method   = "PATCH"
    }

  # restapi_object.cmu-study_study_push_mirror will be created
  + resource "restapi_object" "cmu-study_study_push_mirror" {
      + api_data        = (known after apply)
      + api_response    = (known after apply)
      + create_response = (known after apply)
      + data            = jsonencode(
            {
              + interval       = "8h0m0s"
              + remote_address = "git@github.com:ScottyLabs/study.git"
              + sync_on_commit = true
              + use_ssh        = true
            }
        )
      + id              = (known after apply)
      + id_attribute    = "remote_name"
      + path            = "/api/v1/repos/ScottyLabs/study/push_mirrors"
    }

Plan: 15 to add, 0 to change, 0 to destroy.
╷
│ Error: Unable to read repository
│ 
│   with forgejo_repository.cmu-eats_dining_api,
│   on forgejo_repos.tf.json line 115, in resource.forgejo_repository.cmu-eats_dining_api:
│  115:       },
│ 
│ Repository with ID 65 not found: The target couldn't be found.
╵
╷
│ Error: Unable to read repository
│ 
│   with forgejo_repository.cmu-eats_eats,
│   on forgejo_repos.tf.json line 129, in resource.forgejo_repository.cmu-eats_eats:
│  129:       },
│ 
│ Repository with ID 66 not found: The target couldn't be found.
╵



Plan Summary

10 projects, 8 with changes, 1 with no changes, 1 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 10 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `openbao` dir: `tofu/openbao` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` 1. project: `posthog` dir: `tofu/posthog` workspace: `default` 1. project: `sentry` dir: `tofu/sentry` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 18.98s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_cmu-study_ch0_annadavi will be created + resource "null_resource" "slack_cmu-study_ch0_annadavi" { + id = (known after apply) + triggers = { + "channel" = "C09KN6M3P53" + "user" = "U07M24AKRGV" } } # null_resource.slack_hub_annadavi will be created + resource "null_resource" "slack_hub_annadavi" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U07M24AKRGV" } } # null_resource.slack_leads_annadavi will be created + resource "null_resource" "slack_leads_annadavi" { + id = (known after apply) + triggers = { + "channel" = "C0AN1M0HDV0" + "user" = "U07M24AKRGV" } } Plan: 3 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 3 to add, 0 to change, 0 to destroy. --- ### 2. project: `openbao` dir: `tofu/openbao` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Downloading crates ... Downloaded rustls-webpki v0.103.15 Downloaded rustls v0.23.45 Blocking waiting for file lock on package cache Compiling rustls v0.23.45 Compiling rustls-webpki v0.103.15 Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/215/default/crates/governance-schema) Compiling ureq v3.3.0 Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/215/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/215/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/215/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 18.81s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # vault_identity_group.bus-sign_admins will be updated in-place ! resource "vault_identity_group" "bus-sign_admins" { id = "1b72d8b5-8511-fe92-c98c-2ddb8cbf17a1" name = "bus-sign-admins" ! policies = [ - "bus-sign-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.cal_admins will be updated in-place ! resource "vault_identity_group" "cal_admins" { id = "da96ea82-8537-f0a1-0d42-a7ab1cc3c006" name = "cal-admins" ! policies = [ - "cal-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.changelog_admins will be updated in-place ! resource "vault_identity_group" "changelog_admins" { id = "d73c0ef4-a7d2-d171-ebd1-1c15cee174c8" name = "changelog-admins" ! policies = [ - "changelog-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.cmugpt-agent_admins will be updated in-place ! resource "vault_identity_group" "cmugpt-agent_admins" { id = "cb081cd2-a3cd-81e1-8c20-0914cac5c213" name = "cmugpt-agent-admins" ! policies = [ - "cmugpt-agent-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.cmugpt-mcp-server_admins will be updated in-place ! resource "vault_identity_group" "cmugpt-mcp-server_admins" { id = "a4717a6f-015f-9fae-c6a2-76657b8a7464" name = "cmugpt-mcp-server-admins" ! policies = [ - "cmugpt-mcp-server-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.cmugpt-sms-surface_admins will be updated in-place ! resource "vault_identity_group" "cmugpt-sms-surface_admins" { id = "780f9498-dee7-8cac-9bbc-e1feab5ea2fb" name = "cmugpt-sms-surface-admins" ! policies = [ - "cmugpt-sms-surface-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.cmugpt-surface_admins will be updated in-place ! resource "vault_identity_group" "cmugpt-surface_admins" { id = "3ae18ef3-f370-f163-767a-accfb092ea0d" name = "cmugpt-surface-admins" ! policies = [ - "cmugpt-surface-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.components_admins will be updated in-place ! resource "vault_identity_group" "components_admins" { id = "d2971366-64a4-984e-1e4e-dad3cd07c841" name = "components-admins" ! policies = [ - "components-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.courses_admins will be updated in-place ! resource "vault_identity_group" "courses_admins" { id = "d090ae17-1992-86a5-1f8b-092c937092e8" name = "courses-admins" ! policies = [ - "courses-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.dalmatian_admins will be updated in-place ! resource "vault_identity_group" "dalmatian_admins" { id = "b1d88b18-8b44-c2ea-9220-fc53ee7e3c7e" name = "dalmatian-admins" ! policies = [ - "dalmatian-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.discord-verify_admins will be updated in-place ! resource "vault_identity_group" "discord-verify_admins" { id = "c3e29b3b-0ce5-61a3-3a69-e950f297fa78" name = "discord-verify-admins" ! policies = [ - "discord-verify-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.documentation_admins will be updated in-place ! resource "vault_identity_group" "documentation_admins" { id = "35c5df83-0e17-31fc-7b82-3365d5526dd5" name = "documentation-admins" ! policies = [ - "documentation-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.eats_admins will be updated in-place ! resource "vault_identity_group" "eats_admins" { id = "05959815-c896-9703-32eb-7ba84604303c" name = "eats-admins" ! policies = [ - "eats-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.groupme-mirror_admins will be updated in-place ! resource "vault_identity_group" "groupme-mirror_admins" { id = "e264aabf-8766-67fd-f0bf-860a240ee2c2" name = "groupme-mirror-admins" ! policies = [ - "groupme-mirror-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.housing_admins will be updated in-place ! resource "vault_identity_group" "housing_admins" { id = "7c688f49-59f1-1593-1e19-16f70b86eae2" name = "housing-admins" ! policies = [ - "housing-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.insta_admins will be updated in-place ! resource "vault_identity_group" "insta_admins" { id = "be24e607-e8d5-9fa3-a11a-9a8ca05ede04" name = "insta-admins" ! policies = [ - "insta-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.link-shortener_admins will be updated in-place ! resource "vault_identity_group" "link-shortener_admins" { id = "d3fa8791-4bd4-db20-68b6-79db5ac36a25" name = "link-shortener-admins" ! policies = [ - "link-shortener-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.lost-and-found-v2_admins will be updated in-place ! resource "vault_identity_group" "lost-and-found-v2_admins" { id = "db8b5072-bb86-c9d0-e93f-951c2901bd15" name = "lost-and-found-v2-admins" ! policies = [ - "lost-and-found-v2-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.maps_admins will be updated in-place ! resource "vault_identity_group" "maps_admins" { id = "6b2bd85e-e311-ae83-8a07-30100f0ac12f" name = "maps-admins" ! policies = [ - "maps-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.occ-main_admins will be updated in-place ! resource "vault_identity_group" "occ-main_admins" { id = "4d40db70-1bac-505e-ec8d-5a9d5cf15ac7" name = "occ-main-admins" ! policies = [ - "occ-main-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.quest_admins will be updated in-place ! resource "vault_identity_group" "quest_admins" { id = "c03f2eab-1701-f315-fc17-39814ddcb34b" name = "quest-admins" ! policies = [ - "quest-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.scottylabs-org_admins will be updated in-place ! resource "vault_identity_group" "scottylabs-org_admins" { id = "0dbd914d-b96c-de7b-1ec5-a0ef330b20e0" name = "scottylabs-org-admins" ! policies = [ - "scottylabs-org-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.study_admins will be created + resource "vault_identity_group" "study_admins" { + external_member_entity_ids = false + external_member_group_ids = false + external_policies = false + id = (known after apply) + name = "study-admins" + policies = [ + "study-prod", ] + type = "external" } # vault_identity_group.study_members will be created + resource "vault_identity_group" "study_members" { + external_member_entity_ids = false + external_member_group_ids = false + external_policies = false + id = (known after apply) + name = "study-members" + policies = [ + "study-dev", ] + type = "external" } # vault_identity_group.tartan-vote_admins will be updated in-place ! resource "vault_identity_group" "tartan-vote_admins" { id = "ff77ad01-20b1-c320-37bd-6185b3f98d80" name = "tartan-vote-admins" ! policies = [ - "tartan-vote-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.terrier_admins will be updated in-place ! resource "vault_identity_group" "terrier_admins" { id = "956692a1-84f6-c95c-6d40-2382ead13476" name = "terrier-admins" ! policies = [ - "terrier-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group_alias.study_admins will be created + resource "vault_identity_group_alias" "study_admins" { + canonical_id = (known after apply) + id = (known after apply) + mount_accessor = "auth_oidc_375ac979" + name = "/projects/study/admins" } # vault_identity_group_alias.study_members will be created + resource "vault_identity_group_alias" "study_members" { + canonical_id = (known after apply) + id = (known after apply) + mount_accessor = "auth_oidc_375ac979" + name = "/projects/study" } # vault_policy.study_dev will be created + resource "vault_policy" "study_dev" { + allow_overwrite = (known after apply) + id = (known after apply) + name = "study-dev" + policy = <<-EOT path "secret/data/secretspec/study/dev/*" { capabilities = ["create", "read", "update", "delete", "list"] } path "secret/metadata/secretspec/study/dev/*" { capabilities = ["list", "read", "delete"] } path "secret/data/secretspec/study/preview/*" { capabilities = ["create", "read", "update", "delete", "list"] } path "secret/metadata/secretspec/study/preview/*" { capabilities = ["list", "read", "delete"] } EOT } # vault_policy.study_prod will be created + resource "vault_policy" "study_prod" { + allow_overwrite = (known after apply) + id = (known after apply) + name = "study-prod" + policy = <<-EOT path "secret/data/secretspec/study/prod/*" { capabilities = ["create", "read", "update", "delete", "list"] } path "secret/metadata/secretspec/study/prod/*" { capabilities = ["list", "read", "delete"] } path "secret/data/secretspec/study/staging/*" { capabilities = ["create", "read", "update", "delete", "list"] } path "secret/metadata/secretspec/study/staging/*" { capabilities = ["list", "read", "delete"] } path "secret/data/secretspec/study/ci/*" { capabilities = ["create", "read", "update", "delete", "list"] } path "secret/metadata/secretspec/study/ci/*" { capabilities = ["list", "read", "delete"] } EOT } Plan: 6 to add, 24 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p openbao ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fopenbao%252Fdefault%252Fopenbao) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p openbao ``` Plan: 6 to add, 24 to change, 0 to destroy. --- ### 3. project: `google_play` dir: `tofu/google_play` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 18.80s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` --- ### 4. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 18.90s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_bridge_login_cmu_study will be created + resource "null_resource" "slack_bridge_login_cmu_study" { + id = (known after apply) + triggers = { + "channel" = "C09KN6M3P53" } } # synapse_mautrix_slack_link.cmu_study will be created + resource "synapse_mautrix_slack_link" "cmu_study" { + discord_channel_id = "1512270354943578132" + discord_relay_configured = (known after apply) + id = (known after apply) + room_alias = (known after apply) + room_id = (known after apply) + slack_bridged = (known after apply) + slack_channel_id = "C09KN6M3P53" + team_name = "CMU Study" + team_slug = "cmu-study" } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p matrix_bridges ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fmatrix_bridges%252Fdefault%252Fmatrix_bridges) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p matrix_bridges ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 5. project: `posthog` dir: `tofu/posthog` workspace: `default` <details><summary>Show Output</summary> ```diff Updating crates.io index Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 18.93s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # posthog_project.cmu-study_study will be created + resource "posthog_project" "cmu-study_study" { + api_token = (sensitive value) + id = (known after apply) + name = "study" + organization_id = (known after apply) + timezone = (known after apply) } # vault_kv_secret_v2.cmu-study_study_posthog_host will be created + resource "vault_kv_secret_v2" "cmu-study_study_posthog_host" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/POSTHOG_HOST" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.cmu-study_study_posthog_key will be created + resource "vault_kv_secret_v2" "cmu-study_study_posthog_key" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/POSTHOG_KEY" + path = (known after apply) + custom_metadata (known after apply) } Plan: 3 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p posthog ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fposthog%252Fdefault%252Fposthog) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p posthog ``` Plan: 3 to add, 0 to change, 0 to destroy. --- ### 6. project: `sentry` dir: `tofu/sentry` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 18.98s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # sentry_key.cmu-study_study will be created + resource "sentry_key" "cmu-study_study" { + dsn = (sensitive value) + dsn_csp = (known after apply) + dsn_public = (known after apply) + dsn_secret = (sensitive value) + id = (known after apply) + javascript_loader_script = (known after apply) + name = "default" + organization = "scottylabs" + project = "study" + project_id = (known after apply) + public = (known after apply) + rate_limit_count = (known after apply) + rate_limit_window = (known after apply) + secret = (sensitive value) } # sentry_project.cmu-study_study will be created + resource "sentry_project" "cmu-study_study" { + client_security = (known after apply) + digests_max_delay = (known after apply) + digests_min_delay = (known after apply) + features = (known after apply) + filters = (known after apply) + fingerprinting_rules = (known after apply) + grouping_enhancements = (known after apply) + highlight_tags = (known after apply) + id = (known after apply) + internal_id = (known after apply) + name = "study" + organization = "scottylabs" + resolve_age = (known after apply) + slug = "study" + teams = [ + "cmu-study", ] } # sentry_team.cmu-study will be created + resource "sentry_team" "cmu-study" { + has_access = (known after apply) + id = (known after apply) + internal_id = (known after apply) + is_member = (known after apply) + is_pending = (known after apply) + name = "CMU Study" + organization = "scottylabs" + slug = "cmu-study" + team_id = (known after apply) } # vault_kv_secret_v2.cmu-study_study_sentry_dsn will be created + resource "vault_kv_secret_v2" "cmu-study_study_sentry_dsn" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/SENTRY_DSN" + path = (known after apply) + custom_metadata (known after apply) } Plan: 4 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p sentry ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fsentry%252Fdefault%252Fsentry) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p sentry ``` Plan: 4 to add, 0 to change, 0 to destroy. --- ### 7. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 18.98s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.ops_annadavi will be created + resource "google_cloud_identity_group_membership" "ops_annadavi" { + create_time = (known after apply) + group = "groups/03fwokq036zsp1f" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "annadavi@andrew.cmu.edu" } + roles { + name = "MEMBER" } } # google_cloud_identity_group_membership.tech_annadavi will be created + resource "google_cloud_identity_group_membership" "tech_annadavi" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "annadavi@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 8. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 18.94s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group.project_cmu-study_study will be created + resource "keycloak_group" "project_cmu-study_study" { + id = (known after apply) + name = "study" + parent_id = "9e6890c8-50a0-4b82-86b9-91791fffac25" + path = (known after apply) + realm_id = "scottylabs" } # keycloak_group.project_cmu-study_study_admins will be created + resource "keycloak_group" "project_cmu-study_study_admins" { + id = (known after apply) + name = "admins" + parent_id = (known after apply) + path = (known after apply) + realm_id = "scottylabs" } # keycloak_group_memberships.project_cmu-study_study will be created + resource "keycloak_group_memberships" "project_cmu-study_study" { + group_id = (known after apply) + id = (known after apply) + members = [ + "annadavi", ] + realm_id = "scottylabs" } # keycloak_group_memberships.project_cmu-study_study_admins will be created + resource "keycloak_group_memberships" "project_cmu-study_study_admins" { + group_id = (known after apply) + id = (known after apply) + members = [ + "annadavi", ] + realm_id = "scottylabs" } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "annadavi", # (35 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_openid_client.study will be created + resource "keycloak_openid_client" "study" { + access_token_lifespan = (known after apply) + access_type = "CONFIDENTIAL" + always_display_in_console = false + backchannel_logout_session_required = true + client_authenticator_type = "client-secret" + client_id = "study" + client_offline_session_idle_timeout = (known after apply) + client_offline_session_max_lifespan = (known after apply) + client_secret = (sensitive value) + client_secret_wo = (write-only attribute) + client_session_idle_timeout = (known after apply) + client_session_max_lifespan = (known after apply) + consent_required = (known after apply) + direct_access_grants_enabled = false + display_on_consent_screen = (known after apply) + enabled = true + exclude_issuer_from_auth_response = (known after apply) + exclude_session_state_from_auth_response = (known after apply) + frontchannel_logout_enabled = (known after apply) + full_scope_allowed = true + id = (known after apply) + implicit_flow_enabled = (known after apply) + import = false + name = "study" + oauth2_device_authorization_grant_enabled = false + oauth2_jwt_authorization_grant_enabled = false + realm_id = "scottylabs" + require_dpop_bound_tokens = (known after apply) + resource_server_id = (known after apply) + service_account_user_id = (known after apply) + service_accounts_enabled = (known after apply) + standard_flow_enabled = true + standard_token_exchange_enabled = false + use_refresh_tokens = true + use_refresh_tokens_client_credentials = false + valid_post_logout_redirect_uris = (known after apply) + valid_redirect_uris = [ + "https://oauth.scottylabs.org/oauth2/callback", ] + web_origins = [ + "+", ] } # keycloak_openid_client.study_dev will be created + resource "keycloak_openid_client" "study_dev" { + access_token_lifespan = (known after apply) + access_type = "CONFIDENTIAL" + always_display_in_console = false + backchannel_logout_session_required = true + client_authenticator_type = "client-secret" + client_id = "study-dev" + client_offline_session_idle_timeout = (known after apply) + client_offline_session_max_lifespan = (known after apply) + client_secret = (sensitive value) + client_secret_wo = (write-only attribute) + client_session_idle_timeout = (known after apply) + client_session_max_lifespan = (known after apply) + consent_required = (known after apply) + direct_access_grants_enabled = false + display_on_consent_screen = (known after apply) + enabled = true + exclude_issuer_from_auth_response = (known after apply) + exclude_session_state_from_auth_response = (known after apply) + frontchannel_logout_enabled = (known after apply) + full_scope_allowed = true + id = (known after apply) + implicit_flow_enabled = (known after apply) + import = false + name = "study-dev" + oauth2_device_authorization_grant_enabled = false + oauth2_jwt_authorization_grant_enabled = false + realm_id = "scottylabs" + require_dpop_bound_tokens = (known after apply) + resource_server_id = (known after apply) + service_account_user_id = (known after apply) + service_accounts_enabled = (known after apply) + standard_flow_enabled = true + standard_token_exchange_enabled = false + use_refresh_tokens = true + use_refresh_tokens_client_credentials = false + valid_post_logout_redirect_uris = (known after apply) + valid_redirect_uris = [ + "http://localhost:8090/oauth2/callback", ] + web_origins = [ + "+", ] } # keycloak_openid_client.study_staging will be created + resource "keycloak_openid_client" "study_staging" { + access_token_lifespan = (known after apply) + access_type = "CONFIDENTIAL" + always_display_in_console = false + backchannel_logout_session_required = true + client_authenticator_type = "client-secret" + client_id = "study-staging" + client_offline_session_idle_timeout = (known after apply) + client_offline_session_max_lifespan = (known after apply) + client_secret = (sensitive value) + client_secret_wo = (write-only attribute) + client_session_idle_timeout = (known after apply) + client_session_max_lifespan = (known after apply) + consent_required = (known after apply) + direct_access_grants_enabled = false + display_on_consent_screen = (known after apply) + enabled = true + exclude_issuer_from_auth_response = (known after apply) + exclude_session_state_from_auth_response = (known after apply) + frontchannel_logout_enabled = (known after apply) + full_scope_allowed = true + id = (known after apply) + implicit_flow_enabled = (known after apply) + import = false + name = "study-staging" + oauth2_device_authorization_grant_enabled = false + oauth2_jwt_authorization_grant_enabled = false + realm_id = "scottylabs" + require_dpop_bound_tokens = (known after apply) + resource_server_id = (known after apply) + service_account_user_id = (known after apply) + service_accounts_enabled = (known after apply) + standard_flow_enabled = true + standard_token_exchange_enabled = false + use_refresh_tokens = true + use_refresh_tokens_client_credentials = false + valid_post_logout_redirect_uris = (known after apply) + valid_redirect_uris = [ + "https://oauth.scottylabs.org/oauth2/callback", ] + web_origins = [ + "+", ] } # keycloak_openid_group_membership_protocol_mapper.study will be created + resource "keycloak_openid_group_membership_protocol_mapper" "study" { + add_to_access_token = true + add_to_id_token = true + add_to_token_introspection = false + add_to_userinfo = true + claim_name = "groups" + client_id = (known after apply) + full_path = true + id = (known after apply) + name = "groups" + realm_id = "scottylabs" } # keycloak_openid_group_membership_protocol_mapper.study_dev will be created + resource "keycloak_openid_group_membership_protocol_mapper" "study_dev" { + add_to_access_token = true + add_to_id_token = true + add_to_token_introspection = false + add_to_userinfo = true + claim_name = "groups" + client_id = (known after apply) + full_path = true + id = (known after apply) + name = "groups" + realm_id = "scottylabs" } # keycloak_openid_group_membership_protocol_mapper.study_staging will be created + resource "keycloak_openid_group_membership_protocol_mapper" "study_staging" { + add_to_access_token = true + add_to_id_token = true + add_to_token_introspection = false + add_to_userinfo = true + claim_name = "groups" + client_id = (known after apply) + full_path = true + id = (known after apply) + name = "groups" + realm_id = "scottylabs" } # vault_kv_secret_v2.study_dev_keycloak_realm will be created + resource "vault_kv_secret_v2" "study_dev_keycloak_realm" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/dev/KEYCLOAK_REALM" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_dev_keycloak_url will be created + resource "vault_kv_secret_v2" "study_dev_keycloak_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/dev/KEYCLOAK_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_dev_oauth_relay_url will be created + resource "vault_kv_secret_v2" "study_dev_oauth_relay_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/dev/OAUTH_RELAY_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_dev_oidc_client_id will be created + resource "vault_kv_secret_v2" "study_dev_oidc_client_id" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/dev/OIDC_CLIENT_ID" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_dev_oidc_client_secret will be created + resource "vault_kv_secret_v2" "study_dev_oidc_client_secret" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/dev/OIDC_CLIENT_SECRET" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_dev_project_admin_group will be created + resource "vault_kv_secret_v2" "study_dev_project_admin_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/dev/PROJECT_ADMIN_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_dev_project_group will be created + resource "vault_kv_secret_v2" "study_dev_project_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/dev/PROJECT_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_preview_keycloak_realm will be created + resource "vault_kv_secret_v2" "study_preview_keycloak_realm" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/preview/KEYCLOAK_REALM" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_preview_keycloak_url will be created + resource "vault_kv_secret_v2" "study_preview_keycloak_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/preview/KEYCLOAK_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_preview_oauth_relay_url will be created + resource "vault_kv_secret_v2" "study_preview_oauth_relay_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/preview/OAUTH_RELAY_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_preview_oidc_client_id will be created + resource "vault_kv_secret_v2" "study_preview_oidc_client_id" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/preview/OIDC_CLIENT_ID" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_preview_oidc_client_secret will be created + resource "vault_kv_secret_v2" "study_preview_oidc_client_secret" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/preview/OIDC_CLIENT_SECRET" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_preview_project_admin_group will be created + resource "vault_kv_secret_v2" "study_preview_project_admin_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/preview/PROJECT_ADMIN_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_preview_project_group will be created + resource "vault_kv_secret_v2" "study_preview_project_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/preview/PROJECT_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_prod_keycloak_realm will be created + resource "vault_kv_secret_v2" "study_prod_keycloak_realm" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/KEYCLOAK_REALM" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_prod_keycloak_url will be created + resource "vault_kv_secret_v2" "study_prod_keycloak_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/KEYCLOAK_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_prod_oauth_relay_url will be created + resource "vault_kv_secret_v2" "study_prod_oauth_relay_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/OAUTH_RELAY_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_prod_oidc_client_id will be created + resource "vault_kv_secret_v2" "study_prod_oidc_client_id" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/OIDC_CLIENT_ID" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_prod_oidc_client_secret will be created + resource "vault_kv_secret_v2" "study_prod_oidc_client_secret" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/OIDC_CLIENT_SECRET" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_prod_project_admin_group will be created + resource "vault_kv_secret_v2" "study_prod_project_admin_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/PROJECT_ADMIN_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_prod_project_group will be created + resource "vault_kv_secret_v2" "study_prod_project_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/PROJECT_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_staging_keycloak_realm will be created + resource "vault_kv_secret_v2" "study_staging_keycloak_realm" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/staging/KEYCLOAK_REALM" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_staging_keycloak_url will be created + resource "vault_kv_secret_v2" "study_staging_keycloak_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/staging/KEYCLOAK_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_staging_oauth_relay_url will be created + resource "vault_kv_secret_v2" "study_staging_oauth_relay_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/staging/OAUTH_RELAY_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_staging_oidc_client_id will be created + resource "vault_kv_secret_v2" "study_staging_oidc_client_id" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/staging/OIDC_CLIENT_ID" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_staging_oidc_client_secret will be created + resource "vault_kv_secret_v2" "study_staging_oidc_client_secret" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/staging/OIDC_CLIENT_SECRET" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_staging_project_admin_group will be created + resource "vault_kv_secret_v2" "study_staging_project_admin_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/staging/PROJECT_ADMIN_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_staging_project_group will be created + resource "vault_kv_secret_v2" "study_staging_project_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/staging/PROJECT_GROUP" + path = (known after apply) + custom_metadata (known after apply) } Plan: 38 to add, 1 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 38 to add, 1 to change, 0 to destroy. --- ### 9. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 19.04s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_channel_permission.cmu-study_ch0 will be created + resource "discord_channel_permission" "cmu-study_ch0" { + allow = 1024 + channel_id = "1512270354943578132" + id = (known after apply) + overwrite_id = (known after apply) + type = "role" } # discord_member_roles.annadavi will be created + resource "discord_member_roles" "annadavi" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "991766392590311455" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1519792320755662928" } + role { + has_role = true + role_id = (known after apply) } } # discord_role.cmu-study will be created + resource "discord_role" "cmu-study" { + hoist = false + id = (known after apply) + managed = (known after apply) + mentionable = false + name = "CMU Study" + permissions = 0 + position = (known after apply) + server_id = "1353118209024331948" } Plan: 3 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 3 to add, 0 to change, 0 to destroy. --- ### 10. project: `git` dir: `tofu/git` workspace: `default` **Plan Error** <details><summary>Show Output</summary> ``` running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu plan -input=false -refresh -out /var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git/git-default.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git': exit status 1 Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 18.89s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ data.external.identity_gabrielx: Reading... data.external.identity_arom1a: Reading... data.external.identity_kritdass: Reading... data.external.identity_jc_965: Reading... data.external.identity_tentype: Reading... data.external.identity_gostmeaper: Reading... data.external.identity_jessica_lr_song: Reading... data.external.identity_annadavi: Reading... data.external.identity_jc_965: Read complete after 0s [id=-] data.external.identity_gabrielx: Read complete after 0s [id=-] data.external.identity_jw8: Reading... data.external.identity_xboxbedrock: Reading... data.external.identity_kritdass: Read complete after 0s [id=-] data.external.identity_gostmeaper: Read complete after 0s [id=-] data.external.identity_tentype: Read complete after 0s [id=-] data.external.identity_jiyas: Reading... data.external.identity_rzhu3: Reading... data.external.identity_stringatlas: Reading... data.external.identity_arom1a: Read complete after 0s [id=-] data.external.identity_jessica_lr_song: Read complete after 0s [id=-] data.external.identity_liamwalk: Reading... data.external.identity_thesuperrl: Reading... data.external.identity_annadavi: Read complete after 1s [id=-] data.external.identity_anish: Reading... data.external.identity_stringatlas: Read complete after 1s [id=-] data.external.identity_jw8: Read complete after 1s [id=-] data.external.identity_kluoxx: Reading... data.external.identity_xboxbedrock: Read complete after 1s [id=-] data.external.identity_acchs: Reading... data.external.identity_kothurvidur: Reading... data.external.identity_rzhu3: Read complete after 1s [id=-] data.external.identity_liamwalk: Read complete after 1s [id=-] data.external.identity_dcrodrig: Reading... data.external.identity_thesuperrl: Read complete after 1s [id=-] data.external.identity_rhsen: Reading... data.external.identity_krishsax: Reading... data.external.identity_jiyas: Read complete after 1s [id=-] data.external.identity_amzoeee: Reading... data.external.identity_manrajs: Reading... data.external.identity_maybe_yiyi: Reading... data.external.identity_anish: Read complete after 0s [id=-] data.external.identity_nikhilr: Reading... data.external.identity_kothurvidur: Read complete after 0s [id=-] data.external.identity_acchs: Read complete after 0s [id=-] data.external.identity_dkou: Reading... data.external.identity_brightz: Reading... data.external.identity_dcrodrig: Read complete after 0s [id=-] data.external.identity_krishsax: Read complete after 0s [id=-] data.external.identity_etashjha: Reading... data.external.identity_kluoxx: Read complete after 0s [id=-] data.external.identity_zoexue: Reading... data.external.identity_amzoeee: Read complete after 0s [id=-] data.external.identity_riztofu: Reading... data.external.identity_emmawong: Reading... data.external.identity_rhsen: Read complete after 0s [id=-] data.external.identity_qianxuege: Reading... data.external.identity_manrajs: Read complete after 0s [id=-] data.external.identity_jalenluorion: Reading... data.external.identity_maybe_yiyi: Read complete after 0s [id=-] data.external.identity_kechezon: Reading... data.external.identity_nikhilr: Read complete after 0s [id=-] data.external.identity_siddharthr: Reading... data.external.identity_qianxuege: Read complete after 0s [id=-] data.external.identity_brightz: Read complete after 0s [id=-] data.external.identity_dkou: Read complete after 0s [id=-] random_password.kennel_webhook_secret: Refreshing state... [id=none] forgejo_repository.cmu-insta_insta: Refreshing state... [name=insta] forgejo_repository.devops_collie: Refreshing state... [name=collie] forgejo_repository.uia_scottylabs_org: Refreshing state... [name=scottylabs-org] data.external.identity_etashjha: Read complete after 0s [id=-] data.external.identity_riztofu: Read complete after 0s [id=-] data.external.identity_zoexue: Read complete after 0s [id=-] data.external.identity_emmawong: Read complete after 0s [id=-] forgejo_repository.cbp_discord_verify: Refreshing state... [name=discord-verify] forgejo_repository.devops_documentation: Refreshing state... [name=documentation] data.external.identity_jalenluorion: Read complete after 0s [id=-] forgejo_repository.slai_cmugpt_agent: Refreshing state... [name=cmugpt-agent] forgejo_repository.slai_cmugpt_sms_surface: Refreshing state... [name=cmugpt-sms-surface] forgejo_repository.slai_cmugpt_surface: Refreshing state... [name=cmugpt-surface] data.external.identity_kechezon: Read complete after 0s [id=-] forgejo_repository.lost-and-found_lost_and_found_v2: Refreshing state... [name=lost-and-found-v2] data.external.identity_siddharthr: Read complete after 1s [id=-] forgejo_repository.cmu-housing_housing: Refreshing state... [name=housing] forgejo_repository.cmu-eats_eats: Refreshing state... [name=eats] forgejo_repository.cmu-eats_dining_api: Refreshing state... [name=dining-api] forgejo_repository.slai_mcp_server: Refreshing state... [name=mcp-server] forgejo_repository.devops_kennel: Refreshing state... [name=kennel] forgejo_repository.cbp_bus_sign: Refreshing state... [name=bus-sign] forgejo_repository.cmu-cal_cal: Refreshing state... [name=cal] forgejo_repository.uia_components: Refreshing state... [name=components] forgejo_repository.occ_occ: Refreshing state... [name=occ] forgejo_repository.quest_quest: Refreshing state... [name=quest] forgejo_repository.cbp_groupme_mirror: Refreshing state... [name=groupme-mirror] data.forgejo_organization.this: Reading... data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs] forgejo_repository.cmu-courses_courses: Refreshing state... [name=courses] forgejo_repository.terrier_terrier: Refreshing state... [name=terrier] forgejo_repository.cbp_dalmatian: Refreshing state... [name=dalmatian] forgejo_repository.devops_link_shortener: Refreshing state... [name=link-shortener] forgejo_repository.devops_infrastructure: Refreshing state... [name=infrastructure] forgejo_repository.cmu-maps_maps: Refreshing state... [name=maps] forgejo_repository.cmu-courses_internet_archive: Refreshing state... [name=internet-archive] forgejo_repository.devops_governance: Refreshing state... [name=governance] forgejo_repository.tartan-vote_tartan_vote: Refreshing state... [name=tartan-vote] github_repository.cmu-eats_eats: Refreshing state... [id=eats] github_team.cmu-cal: Refreshing state... [id=15669112] github_repository.cmu-maps_maps: Refreshing state... [id=maps] github_repository.cmu-insta_insta: Refreshing state... [id=insta] github_repository.slai_cmugpt_sms_surface: Refreshing state... [id=cmugpt-sms-surface] github_team.tartan-vote: Refreshing state... [id=17701833] github_repository.slai_mcp_server: Refreshing state... [id=mcp-server] github_team.cmu-eats: Refreshing state... [id=16807600] github_repository.devops_documentation: Refreshing state... [id=documentation] github_team.occ: Refreshing state... [id=19240310] github_repository.uia_scottylabs_org: Refreshing state... [id=scottylabs-org] github_repository.cbp_groupme_mirror: Refreshing state... [id=groupme-mirror] github_repository.cmu-cal_cal: Refreshing state... [id=cal] github_repository.cbp_discord_verify: Refreshing state... [id=discord-verify] github_repository.cmu-eats_dining_api: Refreshing state... [id=dining-api] github_team.cmu-insta: Refreshing state... [id=19240311] github_team.cmu-housing: Refreshing state... [id=17937919] github_repository.devops_kennel: Refreshing state... [id=kennel] github_repository.uia_components: Refreshing state... [id=components] github_repository.devops_infrastructure: Refreshing state... [id=infrastructure] github_repository.cmu-courses_internet_archive: Refreshing state... [id=internet-archive] github_team.slai: Refreshing state... [id=17571328] github_team.cmu-courses: Refreshing state... [id=3643907] github_team.quest: Refreshing state... [id=17330197] github_repository.quest_quest: Refreshing state... [id=quest] github_repository.occ_occ: Refreshing state... [id=occ] github_repository.devops_link_shortener: Refreshing state... [id=link-shortener] github_repository.cmu-courses_courses: Refreshing state... [id=courses] github_team.cmu-maps: Refreshing state... [id=10872360] github_team.devops: Refreshing state... [id=14706218] github_repository.slai_cmugpt_surface: Refreshing state... [id=cmugpt-surface] github_repository.devops_governance: Refreshing state... [id=governance] github_team.terrier: Refreshing state... [id=12028577] github_repository.tartan-vote_tartan_vote: Refreshing state... [id=tartan-vote] github_team.uia: Refreshing state... [id=17529763] github_team.lost-and-found: Refreshing state... [id=4088608] github_repository.terrier_terrier: Refreshing state... [id=terrier] github_repository.slai_cmugpt_agent: Refreshing state... [id=cmugpt-agent] github_repository.devops_collie: Refreshing state... [id=collie] github_team.cbp: Refreshing state... [id=17170099] github_repository.cbp_bus_sign: Refreshing state... [id=bus-sign] github_repository.cmu-housing_housing: Refreshing state... [id=housing] github_repository.cbp_dalmatian: Refreshing state... [id=dalmatian] github_repository.lost-and-found_lost_and_found_v2: Refreshing state... [id=lost-and-found-v2] restapi_object.devops_collie_push_mirror: Refreshing state... [id=remote_mirror_ljvVejNFXaS] restapi_object.cmu-insta_insta_push_mirror: Refreshing state... [id=remote_mirror_891nu6XQ1us] restapi_object.uia_scottylabs_org_push_mirror: Refreshing state... [id=remote_mirror_h6yN6QdS1lD] restapi_object.slai_cmugpt_sms_surface_push_mirror: Refreshing state... [id=remote_mirror__fkBkASN2jA] restapi_object.cbp_discord_verify_push_mirror: Refreshing state... [id=remote_mirror_3MV7Up6cY1X] restapi_object.devops_documentation_push_mirror: Refreshing state... [id=remote_mirror_izmsqaNB6Wx] restapi_object.slai_cmugpt_surface_push_mirror: Refreshing state... [id=remote_mirror_cPTQN6S5SpB] restapi_object.lost-and-found_lost_and_found_v2_push_mirror: Refreshing state... [id=remote_mirror_n_UZBFU-CY4] restapi_object.slai_cmugpt_agent_push_mirror: Refreshing state... [id=remote_mirror_kNY6rVcxdyr] restapi_object.cmu-cal_cal_push_mirror: Refreshing state... [id=remote_mirror_CO243uIONoq] forgejo_team.cmu-maps: Refreshing state... [name=cmu-maps] forgejo_team.courses_leads: Refreshing state... [name=courses-leads] forgejo_team.tartan-vote: Refreshing state... [name=tartan-vote] forgejo_team.cmugpt-sms-surface_leads: Refreshing state... [name=cmugpt-sms-surface-leads] forgejo_team.devops_leads: Refreshing state... [name=devops-leads] forgejo_team.insta_leads: Refreshing state... [name=insta-leads] forgejo_team.uia: Refreshing state... [name=uia] forgejo_team.cmugpt-surface_leads: Refreshing state... [name=cmugpt-surface-leads] forgejo_team.cmu-cal: Refreshing state... [name=cmu-cal] forgejo_team.cbp: Refreshing state... [name=cbp] forgejo_team.cmu-insta: Refreshing state... [name=cmu-insta] forgejo_team.cmu-housing: Refreshing state... [name=cmu-housing] forgejo_team.terrier: Refreshing state... [name=terrier] forgejo_team.components_leads: Refreshing state... [name=components-leads] forgejo_team.link-shortener_leads: Refreshing state... [name=link-shortener-leads] forgejo_team.cmu-courses: Refreshing state... [name=cmu-courses] forgejo_team.housing_leads: Refreshing state... [name=housing-leads] forgejo_team.lost-and-found-v2_leads: Refreshing state... [name=lost-and-found-v2-leads] restapi_object.cmu-housing_housing_push_mirror: Refreshing state... [id=remote_mirror_uoV2HG4aiOT] restapi_object.cmu-housing_housing_docs_webhook: Refreshing state... [id=15] forgejo_team.quest_leads: Refreshing state... [name=quest-leads] forgejo_team.groupme-mirror_leads: Refreshing state... [name=groupme-mirror-leads] forgejo_team.quest: Refreshing state... [name=quest] forgejo_team.dalmatian_leads: Refreshing state... [name=dalmatian-leads] forgejo_team.cmu-eats: Refreshing state... [name=cmu-eats] forgejo_team.cmu-courses_leads: Refreshing state... [name=cmu-courses-leads] forgejo_team.occ-main_leads: Refreshing state... [name=occ-main-leads] forgejo_team.bus-sign_leads: Refreshing state... [name=bus-sign-leads] forgejo_team.cmugpt-mcp-server_leads: Refreshing state... [name=cmugpt-mcp-server-leads] forgejo_team.tartan-vote_leads: Refreshing state... [name=tartan-vote-leads] forgejo_team.cal_leads: Refreshing state... [name=cal-leads] forgejo_team.scottylabs-org_leads: Refreshing state... [name=scottylabs-org-leads] forgejo_team.cmugpt-agent_leads: Refreshing state... [name=cmugpt-agent-leads] forgejo_team.eats_leads: Refreshing state... [name=eats-leads] forgejo_team.devops: Refreshing state... [name=devops] forgejo_team.occ: Refreshing state... [name=occ] forgejo_team.terrier_leads: Refreshing state... [name=terrier-leads] forgejo_team.slai: Refreshing state... [name=slai] forgejo_team.discord-verify_leads: Refreshing state... [name=discord-verify-leads] forgejo_team.lost-and-found: Refreshing state... [name=lost-and-found] forgejo_team.maps_leads: Refreshing state... [name=maps-leads] forgejo_team.documentation_leads: Refreshing state... [name=documentation-leads] restapi_object.slai_mcp_server_push_mirror: Refreshing state... [id=remote_mirror_Tqh2y0R_vDp] restapi_object.devops_kennel_push_mirror: Refreshing state... [id=remote_mirror_cT9psJkhTQ4] restapi_object.cbp_bus_sign_push_mirror: Refreshing state... [id=remote_mirror_MkKxQ2vA9pc] restapi_object.quest_quest_push_mirror: Refreshing state... [id=remote_mirror_U9d9BPAon29] restapi_object.uia_components_push_mirror: Refreshing state... [id=remote_mirror_Kwwa8TDnTgc] restapi_object.occ_occ_push_mirror: Refreshing state... [id=remote_mirror_moBLq1YG1qz] restapi_object.cbp_groupme_mirror_push_mirror: Refreshing state... [id=remote_mirror_U61HJHB9VGE] restapi_object.cmu-courses_courses_push_mirror: Refreshing state... [id=remote_mirror_WMR9yQwKFDF] restapi_object.terrier_terrier_push_mirror: Refreshing state... [id=remote_mirror_YopDgZksZv5] restapi_object.devops_link_shortener_push_mirror: Refreshing state... [id=remote_mirror_cp8daMpt82v] restapi_object.cbp_dalmatian_push_mirror: Refreshing state... [id=remote_mirror_qzlCpLK153F] restapi_object.cbp_dalmatian_docs_webhook: Refreshing state... [id=7] restapi_object.devops_infrastructure_push_mirror: Refreshing state... [id=remote_mirror_oYtFpKVzTaZ] restapi_object.cmu-courses_internet_archive_push_mirror: Refreshing state... [id=remote_mirror_58X__Yfs0bR] restapi_object.devops_governance_push_mirror: Refreshing state... [id=remote_mirror_lt9JzkX1jMq] restapi_object.cmu-maps_maps_push_mirror: Refreshing state... [id=remote_mirror_e4_FUnvoslJ] restapi_object.tartan-vote_tartan_vote_push_mirror: Refreshing state... [id=remote_mirror_EUFq4sFSnH5] restapi_object.tartan-vote_tartan_vote_docs_webhook: Refreshing state... [id=25] github_team_membership.cmu-cal_anish: Refreshing state... [id=15669112:ap-1] github_team_membership.cmu-cal_jalenluorion: Refreshing state... [id=15669112:jalenluorion] github_actions_repository_permissions.slai_cmugpt_sms_surface_actions: Refreshing state... [id=cmugpt-sms-surface] github_team_membership.tartan-vote_xboxbedrock: Refreshing state... [id=17701833:xboxbedrock] github_team_membership.tartan-vote_anish: Refreshing state... [id=17701833:ap-1] github_team_membership.tartan-vote_riztofu: Refreshing state... [id=17701833:riztofu] github_team_membership.tartan-vote_tentype: Refreshing state... [id=17701833:tentype] github_team_membership.tartan-vote_maybe_yiyi: Refreshing state... [id=17701833:maybe-yiyi] github_team_membership.cmu-eats_arom1a: Refreshing state... [id=16807600:arom1a] github_team_membership.occ_xboxbedrock: Refreshing state... [id=19240310:xboxbedrock] github_team_membership.occ_dcrodrig: Refreshing state... [id=19240310:newhacker1746] github_team_membership.occ_anish: Refreshing state... [id=19240310:ap-1] github_team_membership.occ_acchs: Refreshing state... [id=19240310:acchs-at-cmu] github_actions_repository_permissions.cmu-eats_eats_actions: Refreshing state... [id=eats] github_actions_repository_permissions.cmu-maps_maps_actions: Refreshing state... [id=maps] github_actions_repository_permissions.cmu-insta_insta_actions: Refreshing state... [id=insta] github_actions_repository_permissions.cbp_groupme_mirror_actions: Refreshing state... [id=groupme-mirror] github_actions_repository_permissions.slai_mcp_server_actions: Refreshing state... [id=mcp-server] github_actions_repository_permissions.cbp_discord_verify_actions: Refreshing state... [id=discord-verify] github_actions_repository_permissions.devops_documentation_actions: Refreshing state... [id=documentation] github_team_membership.cmu-insta_etashjha: Refreshing state... [id=19240311:etashj] github_actions_repository_permissions.uia_scottylabs_org_actions: Refreshing state... [id=scottylabs-org] github_team_membership.cmu-housing_emmawong: Refreshing state... [id=17937919:emw868] github_team_membership.cmu-housing_anish: Refreshing state... [id=17937919:ap-1] github_team_membership.cmu-housing_nikhilr: Refreshing state... [id=17937919:ecstaticpilot] github_team_membership.cmu-housing_tentype: Refreshing state... [id=17937919:tentype] github_team_membership.cmu-housing_zoexue: Refreshing state... [id=17937919:o-zox-o] github_team_membership.cmu-housing_xboxbedrock: Refreshing state... [id=17937919:xboxbedrock] github_team_membership.cmu-housing_gostmeaper: Refreshing state... [id=17937919:gostmeaper] github_team_membership.cmu-housing_jessica_lr_song: Refreshing state... [id=17937919:jessica-lr-song] github_team_membership.cmu-housing_manrajs: Refreshing state... [id=17937919:manraj10] github_team_membership.cmu-housing_amzoeee: Refreshing state... [id=17937919:amzoeee] github_team_membership.cmu-housing_kluoxx: Refreshing state... [id=17937919:luokaden888] github_team_membership.cmu-housing_rhsen: Refreshing state... [id=17937919:rohansen08] github_team_membership.cmu-housing_stringatlas: Refreshing state... [id=17937919:stringatlas] github_actions_repository_permissions.uia_components_actions: Refreshing state... [id=components] github_actions_repository_permissions.cmu-cal_cal_actions: Refreshing state... [id=cal] github_actions_repository_permissions.cmu-courses_internet_archive_actions: Refreshing state... [id=internet-archive] github_team_membership.slai_jc_965: Refreshing state... [id=17571328:jc-965] github_team_membership.slai_anish: Refreshing state... [id=17571328:ap-1] github_team_membership.slai_jiyas: Refreshing state... [id=17571328:jiya2301] github_team_membership.slai_xboxbedrock: Refreshing state... [id=17571328:xboxbedrock] github_team_membership.slai_qianxuege: Refreshing state... [id=17571328:qianxuege] github_team_membership.slai_krishsax: Refreshing state... [id=17571328:krishsaxena] github_team_membership.slai_rzhu3: Refreshing state... [id=17571328:richard-31415] github_actions_repository_permissions.cmu-eats_dining_api_actions: Refreshing state... [id=dining-api] github_team_membership.cmu-courses_anish: Refreshing state... [id=3643907:ap-1] github_team_membership.cmu-courses_etashjha: Refreshing state... [id=3643907:etashj] github_team_membership.cmu-courses_dkou: Refreshing state... [id=3643907:sharkree] github_team_membership.cmu-courses_gabrielx: Refreshing state... [id=3643907:gabriel-xu] github_team_membership.quest_xboxbedrock: Refreshing state... [id=17330197:xboxbedrock] github_team_membership.quest_brightz: Refreshing state... [id=17330197:bright-zheng] github_team_membership.quest_jw8: Refreshing state... [id=17330197:jeffreyyijiwang] github_team_membership.quest_kechezon: Refreshing state... [id=17330197:kechezon] github_team_membership.quest_anish: Refreshing state... [id=17330197:ap-1] github_actions_repository_permissions.devops_kennel_actions: Refreshing state... [id=kennel] github_actions_repository_permissions.devops_infrastructure_actions: Refreshing state... [id=infrastructure] github_actions_repository_permissions.cmu-courses_courses_actions: Refreshing state... [id=courses] github_team_membership.cmu-maps_anish: Refreshing state... [id=10872360:ap-1] github_team_membership.cmu-maps_siddharthr: Refreshing state... [id=10872360:overthrowing] github_team_membership.devops_anish: Refreshing state... [id=14706218:ap-1] github_team_membership.devops_thesuperrl: Refreshing state... [id=14706218:thesuperrl] github_team_membership.devops_xboxbedrock: Refreshing state... [id=14706218:xboxbedrock] github_actions_repository_permissions.slai_cmugpt_surface_actions: Refreshing state... [id=cmugpt-surface] github_actions_repository_permissions.quest_quest_actions: Refreshing state... [id=quest] github_team_membership.terrier_kothurvidur: Refreshing state... [id=12028577:kothurvidur] github_team_membership.terrier_dkou: Refreshing state... [id=12028577:sharkree] github_team_membership.terrier_kritdass: Refreshing state... [id=12028577:kritdass] github_team_membership.terrier_anish: Refreshing state... [id=12028577:ap-1] github_team_membership.terrier_amzoeee: Refreshing state... [id=12028577:amzoeee] github_team_membership.terrier_etashjha: Refreshing state... [id=12028577:etashj] github_actions_repository_permissions.occ_occ_actions: Refreshing state... [id=occ] github_actions_repository_permissions.devops_link_shortener_actions: Refreshing state... [id=link-shortener] github_team_membership.uia_anish: Refreshing state... [id=17529763:ap-1] github_team_membership.lost-and-found_anish: Refreshing state... [id=4088608:ap-1] github_actions_repository_permissions.slai_cmugpt_agent_actions: Refreshing state... [id=cmugpt-agent] restapi_object.slai_mcp_server_kennel_webhook: Refreshing state... [id=13] restapi_object.uia_components_kennel_webhook: Refreshing state... [id=27] restapi_object.devops_collie_kennel_webhook: Refreshing state... [id=17] restapi_object.slai_cmugpt_sms_surface_kennel_webhook: Refreshing state... [id=19] restapi_object.tartan-vote_tartan_vote_kennel_webhook: Refreshing state... [id=24] restapi_object.cbp_discord_verify_kennel_webhook: Refreshing state... [id=18] restapi_object.quest_quest_kennel_webhook: Refreshing state... [id=10] restapi_object.devops_kennel_kennel_webhook: Refreshing state... [id=14] restapi_object.cbp_bus_sign_kennel_webhook: Refreshing state... [id=21] restapi_object.cmu-maps_maps_kennel_webhook: Refreshing state... [id=9] restapi_object.cbp_dalmatian_kennel_webhook: Refreshing state... [id=6] restapi_object.uia_scottylabs_org_kennel_webhook: Refreshing state... [id=29] restapi_object.devops_governance_kennel_webhook: Refreshing state... [id=22] restapi_object.cmu-cal_cal_kennel_webhook: Refreshing state... [id=20] restapi_object.cmu-courses_courses_kennel_webhook: Refreshing state... [id=11] restapi_object.slai_cmugpt_surface_kennel_webhook: Refreshing state... [id=23] restapi_object.cmu-housing_housing_kennel_webhook: Refreshing state... [id=16] restapi_object.terrier_terrier_kennel_webhook: Refreshing state... [id=12] restapi_object.slai_cmugpt_agent_kennel_webhook: Refreshing state... [id=28] restapi_object.devops_link_shortener_kennel_webhook: Refreshing state... [id=8] restapi_object.devops_infrastructure_kennel_webhook: Refreshing state... [id=26] github_repository_deploy_key.cmu-insta_insta_mirror_deploy_key: Refreshing state... [id=insta:161720245] github_repository_deploy_key.slai_cmugpt_sms_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-sms-surface:159665365] github_repository_deploy_key.uia_scottylabs_org_mirror_deploy_key: Refreshing state... [id=scottylabs-org:159665361] github_repository_deploy_key.cbp_discord_verify_mirror_deploy_key: Refreshing state... [id=discord-verify:159665340] github_actions_repository_permissions.devops_governance_actions: Refreshing state... [id=governance] github_repository_deploy_key.devops_documentation_mirror_deploy_key: Refreshing state... [id=documentation:159665362] github_repository_deploy_key.slai_cmugpt_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-surface:159665348] github_repository_deploy_key.slai_cmugpt_agent_mirror_deploy_key: Refreshing state... [id=cmugpt-agent:159665360] github_repository_deploy_key.cmu-cal_cal_mirror_deploy_key: Refreshing state... [id=cal:159665344] forgejo_team_member.cmu-maps_siddharthr: Refreshing state... forgejo_team_member.cmu-maps_anish: Refreshing state... forgejo_team_repository.cmu-maps_maps: Refreshing state... forgejo_team_member.tartan-vote_anish: Refreshing state... forgejo_team_member.tartan-vote_maybe_yiyi: Refreshing state... forgejo_team_repository.tartan-vote_tartan_vote: Refreshing state... forgejo_team_member.tartan-vote_xboxbedrock: Refreshing state... forgejo_team_member.courses_leads_anish: Refreshing state... forgejo_team_member.courses_leads_etashjha: Refreshing state... forgejo_team_member.tartan-vote_tentype: Refreshing state... forgejo_team_member.tartan-vote_riztofu: Refreshing state... forgejo_team_member.devops_leads_thesuperrl: Refreshing state... forgejo_team_member.devops_leads_anish: Refreshing state... forgejo_team_member.cmugpt-sms-surface_leads_krishsax: Refreshing state... forgejo_team_member.cmugpt-sms-surface_leads_anish: Refreshing state... forgejo_team_member.insta_leads_etashjha: Refreshing state... forgejo_team_member.uia_anish: Refreshing state... forgejo_team_repository.uia_scottylabs_org: Refreshing state... forgejo_team_repository.uia_components: Refreshing state... forgejo_team_member.cmugpt-surface_leads_anish: Refreshing state... forgejo_team_member.cmugpt-surface_leads_krishsax: Refreshing state... forgejo_team_member.cmu-cal_anish: Refreshing state... forgejo_team_member.cmu-cal_jalenluorion: Refreshing state... forgejo_team_repository.cmu-cal_cal: Refreshing state... forgejo_team_member.cmu-insta_etashjha: Refreshing state... forgejo_team_repository.cmu-insta_insta: Refreshing state... forgejo_team_member.cbp_tentype: Refreshing state... forgejo_team_member.cbp_xboxbedrock: Refreshing state... forgejo_team_member.cbp_anish: Refreshing state... forgejo_team_member.cbp_liamwalk: Refreshing state... forgejo_team_repository.cbp_discord_verify: Refreshing state... forgejo_team_member.cbp_maybe_yiyi: Refreshing state... forgejo_team_repository.cbp_dalmatian: Refreshing state... forgejo_team_member.cbp_nikhilr: Refreshing state... forgejo_team_repository.cbp_bus_sign: Refreshing state... forgejo_team_repository.cbp_groupme_mirror: Refreshing state... forgejo_team_member.cmu-housing_tentype: Refreshing state... forgejo_team_member.terrier_amzoeee: Refreshing state... forgejo_team_member.cmu-housing_zoexue: Refreshing state... forgejo_team_member.terrier_anish: Refreshing state... forgejo_team_member.cmu-housing_kluoxx: Refreshing state... forgejo_team_member.cmu-housing_anish: Refreshing state... forgejo_team_member.cmu-housing_stringatlas: Refreshing state... forgejo_team_member.cmu-housing_rhsen: Refreshing state... forgejo_team_member.terrier_kothurvidur: Refreshing state... forgejo_team_member.terrier_kritdass: Refreshing state... forgejo_team_repository.terrier_terrier: Refreshing state... forgejo_team_member.cmu-housing_xboxbedrock: Refreshing state... forgejo_team_member.terrier_etashjha: Refreshing state... forgejo_team_member.cmu-housing_nikhilr: Refreshing state... forgejo_team_member.cmu-housing_emmawong: Refreshing state... forgejo_team_repository.cmu-housing_housing: Refreshing state... forgejo_team_member.cmu-housing_jessica_lr_song: Refreshing state... forgejo_team_member.cmu-housing_amzoeee: Refreshing state... forgejo_team_member.cmu-housing_manrajs: Refreshing state... forgejo_team_member.cmu-housing_gostmeaper: Refreshing state... forgejo_team_member.terrier_dkou: Refreshing state... forgejo_team_member.link-shortener_leads_thesuperrl: Refreshing state... forgejo_team_member.link-shortener_leads_anish: Refreshing state... forgejo_team_repository.cmu-courses_internet_archive: Refreshing state... forgejo_team_repository.cmu-courses_courses: Refreshing state... forgejo_team_member.cmu-courses_etashjha: Refreshing state... forgejo_team_member.cmu-courses_dkou: Refreshing state... forgejo_team_member.cmu-courses_anish: Refreshing state... forgejo_team_member.cmu-courses_gabrielx: Refreshing state... forgejo_team_member.components_leads_anish: Refreshing state... forgejo_team_member.housing_leads_tentype: Refreshing state... forgejo_team_member.housing_leads_gostmeaper: Refreshing state... forgejo_team_member.housing_leads_emmawong: Refreshing state... forgejo_team_member.housing_leads_anish: Refreshing state... forgejo_team_member.housing_leads_nikhilr: Refreshing state... forgejo_team_member.lost-and-found-v2_leads_anish: Refreshing state... github_actions_repository_permissions.devops_collie_actions: Refreshing state... [id=collie] github_repository_deploy_key.devops_collie_mirror_deploy_key: Refreshing state... [id=collie:159665332] forgejo_team_member.quest_leads_xboxbedrock: Refreshing state... forgejo_team_member.quest_leads_anish: Refreshing state... forgejo_team_member.groupme-mirror_leads_anish: Refreshing state... forgejo_team_member.groupme-mirror_leads_maybe_yiyi: Refreshing state... forgejo_team_repository.quest_quest: Refreshing state... forgejo_team_member.groupme-mirror_leads_tentype: Refreshing state... forgejo_team_member.quest_kechezon: Refreshing state... forgejo_team_member.quest_jw8: Refreshing state... forgejo_team_member.quest_brightz: Refreshing state... forgejo_team_member.quest_anish: Refreshing state... forgejo_team_member.quest_xboxbedrock: Refreshing state... forgejo_team_member.dalmatian_leads_tentype: Refreshing state... forgejo_team_member.dalmatian_leads_anish: Refreshing state... forgejo_team_member.dalmatian_leads_maybe_yiyi: Refreshing state... forgejo_team_member.cmu-eats_arom1a: Refreshing state... forgejo_team_member.cmu-courses_leads_etashjha: Refreshing state... forgejo_team_member.cmu-courses_leads_anish: Refreshing state... forgejo_team_member.occ-main_leads_dcrodrig: Refreshing state... forgejo_team_member.occ-main_leads_anish: Refreshing state... forgejo_team_member.bus-sign_leads_maybe_yiyi: Refreshing state... forgejo_team_member.bus-sign_leads_tentype: Refreshing state... forgejo_team_member.bus-sign_leads_anish: Refreshing state... forgejo_team_member.cmugpt-mcp-server_leads_anish: Refreshing state... forgejo_team_member.cmugpt-mcp-server_leads_krishsax: Refreshing state... forgejo_team_member.tartan-vote_leads_maybe_yiyi: Refreshing state... forgejo_team_member.tartan-vote_leads_anish: Refreshing state... forgejo_team_member.scottylabs-org_leads_anish: Refreshing state... forgejo_team_member.cal_leads_jalenluorion: Refreshing state... forgejo_team_member.cal_leads_anish: Refreshing state... forgejo_team_member.cmugpt-agent_leads_anish: Refreshing state... forgejo_team_member.cmugpt-agent_leads_krishsax: Refreshing state... github_team_membership.cbp_liamwalk: Refreshing state... [id=17170099:pandaaured] github_team_membership.cbp_nikhilr: Refreshing state... [id=17170099:ecstaticpilot] github_team_membership.cbp_anish: Refreshing state... [id=17170099:ap-1] github_team_membership.cbp_tentype: Refreshing state... [id=17170099:tentype] github_team_membership.cbp_maybe_yiyi: Refreshing state... [id=17170099:maybe-yiyi] github_team_membership.cbp_xboxbedrock: Refreshing state... [id=17170099:xboxbedrock] forgejo_team_repository.devops_link_shortener: Refreshing state... forgejo_team_repository.devops_kennel: Refreshing state... forgejo_team_repository.devops_governance: Refreshing state... forgejo_team_repository.devops_infrastructure: Refreshing state... forgejo_team_member.devops_thesuperrl: Refreshing state... forgejo_team_member.devops_anish: Refreshing state... forgejo_team_member.devops_xboxbedrock: Refreshing state... forgejo_team_member.eats_leads_arom1a: Refreshing state... forgejo_team_repository.devops_documentation: Refreshing state... forgejo_team_repository.devops_collie: Refreshing state... forgejo_team_member.slai_anish: Refreshing state... forgejo_team_member.slai_jiyas: Refreshing state... forgejo_team_repository.slai_cmugpt_sms_surface: Refreshing state... forgejo_team_repository.slai_mcp_server: Refreshing state... forgejo_team_member.slai_jc_965: Refreshing state... forgejo_team_member.slai_krishsax: Refreshing state... forgejo_team_member.slai_qianxuege: Refreshing state... forgejo_team_member.slai_rzhu3: Refreshing state... forgejo_team_member.slai_xboxbedrock: Refreshing state... forgejo_team_repository.slai_cmugpt_surface: Refreshing state... forgejo_team_member.occ_xboxbedrock: Refreshing state... forgejo_team_member.occ_dcrodrig: Refreshing state... forgejo_team_repository.occ_occ: Refreshing state... forgejo_team_member.occ_acchs: Refreshing state... forgejo_team_member.occ_anish: Refreshing state... forgejo_team_member.terrier_leads_kritdass: Refreshing state... forgejo_team_member.terrier_leads_anish: Refreshing state... forgejo_team_repository.slai_cmugpt_agent: Refreshing state... forgejo_team_member.discord-verify_leads_tentype: Refreshing state... forgejo_team_member.discord-verify_leads_maybe_yiyi: Refreshing state... forgejo_team_member.discord-verify_leads_anish: Refreshing state... forgejo_team_repository.lost-and-found_lost_and_found_v2: Refreshing state... forgejo_team_member.lost-and-found_anish: Refreshing state... forgejo_team_member.maps_leads_siddharthr: Refreshing state... forgejo_team_member.maps_leads_anish: Refreshing state... forgejo_team_member.documentation_leads_thesuperrl: Refreshing state... forgejo_team_member.documentation_leads_anish: Refreshing state... github_repository_deploy_key.devops_kennel_mirror_deploy_key: Refreshing state... [id=kennel:159665328] github_repository_deploy_key.slai_mcp_server_mirror_deploy_key: Refreshing state... [id=mcp-server:159665358] github_repository_deploy_key.quest_quest_mirror_deploy_key: Refreshing state... [id=quest:159665323] github_repository_deploy_key.occ_occ_mirror_deploy_key: Refreshing state... [id=occ:161720242] github_repository_deploy_key.uia_components_mirror_deploy_key: Refreshing state... [id=components:159665353] github_repository_deploy_key.cbp_groupme_mirror_mirror_deploy_key: Refreshing state... [id=groupme-mirror:159665366] github_repository_deploy_key.cmu-courses_courses_mirror_deploy_key: Refreshing state... [id=courses:159665327] github_repository_deploy_key.devops_link_shortener_mirror_deploy_key: Refreshing state... [id=link-shortener:159665345] github_repository_deploy_key.devops_infrastructure_mirror_deploy_key: Refreshing state... [id=infrastructure:159665343] github_repository_deploy_key.cmu-courses_internet_archive_mirror_deploy_key: Refreshing state... [id=internet-archive:159665356] github_repository_deploy_key.devops_governance_mirror_deploy_key: Refreshing state... [id=governance:159665346] github_repository_deploy_key.cmu-maps_maps_mirror_deploy_key: Refreshing state... [id=maps:159665325] github_actions_repository_permissions.tartan-vote_tartan_vote_actions: Refreshing state... [id=tartan-vote] github_repository_deploy_key.tartan-vote_tartan_vote_mirror_deploy_key: Refreshing state... [id=tartan-vote:159665350] github_actions_repository_permissions.lost-and-found_lost_and_found_v2_actions: Refreshing state... [id=lost-and-found-v2] github_repository_deploy_key.lost-and-found_lost_and_found_v2_mirror_deploy_key: Refreshing state... [id=lost-and-found-v2:159665337] github_repository_deploy_key.terrier_terrier_mirror_deploy_key: Refreshing state... [id=terrier:159665339] github_actions_repository_permissions.terrier_terrier_actions: Refreshing state... [id=terrier] github_repository_deploy_key.cbp_bus_sign_mirror_deploy_key: Refreshing state... [id=bus-sign:159665342] github_actions_repository_permissions.cbp_bus_sign_actions: Refreshing state... [id=bus-sign] github_actions_repository_permissions.cmu-housing_housing_actions: Refreshing state... [id=housing] github_repository_deploy_key.cmu-housing_housing_mirror_deploy_key: Refreshing state... [id=housing:159665369] github_repository_deploy_key.cbp_dalmatian_mirror_deploy_key: Refreshing state... [id=dalmatian:159665367] github_actions_repository_permissions.cbp_dalmatian_actions: Refreshing state... [id=dalmatian] forgejo_team_repository.maps_leads_maps: Refreshing state... forgejo_team_repository.tartan-vote_leads_tartan_vote: Refreshing state... github_repository_ruleset.cmu-insta_insta_ruleset: Refreshing state... [id=21841636] forgejo_team_repository.scottylabs-org_leads_scottylabs_org: Refreshing state... forgejo_team_repository.components_leads_components: Refreshing state... forgejo_team_repository.cal_leads_cal: Refreshing state... github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Refreshing state... [id=17653622] forgejo_team_repository.insta_leads_insta: Refreshing state... forgejo_team_repository.discord-verify_leads_discord_verify: Refreshing state... forgejo_team_repository.dalmatian_leads_dalmatian: Refreshing state... forgejo_team_repository.bus-sign_leads_bus_sign: Refreshing state... forgejo_team_repository.groupme-mirror_leads_groupme_mirror: Refreshing state... github_repository_ruleset.uia_scottylabs_org_ruleset: Refreshing state... [id=18112214] forgejo_team_repository.terrier_leads_terrier: Refreshing state... forgejo_team_repository.housing_leads_housing: Refreshing state... forgejo_team_repository.courses_leads_courses: Refreshing state... forgejo_team_repository.cmu-courses_leads_internet_archive: Refreshing state... github_repository_ruleset.cbp_discord_verify_ruleset: Refreshing state... [id=11449154] forgejo_team_repository.quest_leads_quest: Refreshing state... github_repository_ruleset.devops_documentation_ruleset: Refreshing state... [id=17079368] github_repository_ruleset.slai_cmugpt_surface_ruleset: Refreshing state... [id=16476662] github_repository_ruleset.slai_cmugpt_agent_ruleset: Refreshing state... [id=16476661] forgejo_team_repository.link-shortener_leads_link_shortener: Refreshing state... forgejo_team_repository.devops_leads_kennel: Refreshing state... forgejo_team_repository.devops_leads_governance: Refreshing state... forgejo_team_repository.devops_leads_infrastructure: Refreshing state... github_repository_ruleset.cmu-cal_cal_ruleset: Refreshing state... [id=18919281] forgejo_team_repository.documentation_leads_documentation: Refreshing state... forgejo_team_repository.devops_leads_collie: Refreshing state... forgejo_team_repository.cmugpt-sms-surface_leads_cmugpt_sms_surface: Refreshing state... forgejo_team_repository.cmugpt-mcp-server_leads_mcp_server: Refreshing state... forgejo_team_repository.cmugpt-surface_leads_cmugpt_surface: Refreshing state... forgejo_team_repository.occ-main_leads_occ: Refreshing state... forgejo_team_repository.cmugpt-agent_leads_cmugpt_agent: Refreshing state... forgejo_team_repository.lost-and-found-v2_leads_lost_and_found_v2: Refreshing state... github_repository_ruleset.devops_collie_ruleset: Refreshing state... [id=18865204] github_repository_ruleset.devops_kennel_ruleset: Refreshing state... [id=13193453] github_repository_ruleset.slai_mcp_server_ruleset: Refreshing state... [id=17082154] github_repository_ruleset.quest_quest_ruleset: Refreshing state... [id=15750052] github_repository_ruleset.occ_occ_ruleset: Refreshing state... [id=21841634] github_repository_ruleset.uia_components_ruleset: Refreshing state... [id=16366085] github_repository_ruleset.cbp_groupme_mirror_ruleset: Refreshing state... [id=12503296] github_repository_ruleset.cmu-courses_courses_ruleset: Refreshing state... [id=15855161] github_repository_ruleset.devops_link_shortener_ruleset: Refreshing state... [id=17911647] github_repository_ruleset.devops_infrastructure_ruleset: Refreshing state... [id=11275118] github_repository_ruleset.cmu-courses_internet_archive_ruleset: Refreshing state... [id=11988074] github_repository_ruleset.devops_governance_ruleset: Refreshing state... [id=14488792] github_repository_ruleset.cmu-maps_maps_ruleset: Refreshing state... [id=18864961] github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Refreshing state... [id=16848966] github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Refreshing state... [id=18920618] github_repository_ruleset.terrier_terrier_ruleset: Refreshing state... [id=12671003] github_repository_ruleset.cbp_bus_sign_ruleset: Refreshing state... [id=15236550] github_repository_ruleset.cmu-housing_housing_ruleset: Refreshing state... [id=17421604] github_repository_ruleset.cbp_dalmatian_ruleset: Refreshing state... [id=17238915] OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu planned the following actions, but then encountered a problem: # forgejo_repository.cmu-study_study will be created + resource "forgejo_repository" "cmu-study_study" { + allow_manual_merge = false + allow_merge_commits = true + allow_rebase = true + allow_rebase_explicit = true + allow_squash_merge = true + archive_on_destroy = false + archived = false + auto_init = true + autodetect_manual_merge = false + avatar_url = (known after apply) + clone_url = (known after apply) + created_at = (known after apply) + default_branch = "main" + default_merge_style = "merge" + description = "A platform for creating and joining study groups" + empty = (known after apply) + external_tracker = (known after apply) + external_wiki = (known after apply) + fork = (known after apply) + forks_count = (known after apply) + full_name = (known after apply) + has_actions = true + has_issues = true + has_packages = true + has_projects = true + has_pull_requests = true + has_releases = true + has_wiki = true + html_url = (known after apply) + id = (known after apply) + ignore_whitespace_conflicts = false + internal = (known after apply) + internal_tracker = (known after apply) + labels = false + lfs = false + milestones = false + mirror = false + mirror_interval = (known after apply) + mirror_updated = (known after apply) + name = "study" + open_issues_count = (known after apply) + open_pr_counter = (known after apply) + owner = "ScottyLabs" + parent_id = (known after apply) + permissions = (known after apply) + private = false + release_counter = (known after apply) + size = (known after apply) + ssh_url = (known after apply) + stars_count = (known after apply) + template = false + trust_model = "default" + updated_at = (known after apply) + watchers_count = (known after apply) + website = "https://cmustudy.com" } # forgejo_team.cmu-study will be created + resource "forgejo_team" "cmu-study" { + can_create_org_repo = (known after apply) + description = "CMU Study" + id = (known after apply) + includes_all_repositories = (known after apply) + name = "cmu-study" + organization = (known after apply) + organization_id = 3 + permission = "write" + units_map = { + "repo.code" = "write" + "repo.issues" = "write" + "repo.pulls" = "write" } } # forgejo_team.study_leads will be created + resource "forgejo_team" "study_leads" { + can_create_org_repo = false + description = "CMU Study leads" + id = (known after apply) + includes_all_repositories = false + name = "study-leads" + organization = (known after apply) + organization_id = 3 + permission = "admin" + units_map = { + "repo.actions" = "admin" + "repo.code" = "admin" + "repo.ext_issues" = "admin" + "repo.ext_wiki" = "admin" + "repo.issues" = "admin" + "repo.packages" = "admin" + "repo.projects" = "admin" + "repo.pulls" = "admin" + "repo.releases" = "admin" + "repo.wiki" = "admin" } } # forgejo_team_member.cmu-study_annadavi will be created + resource "forgejo_team_member" "cmu-study_annadavi" { + team_id = (known after apply) + user = "annadavi" } # forgejo_team_member.study_leads_annadavi will be created + resource "forgejo_team_member" "study_leads_annadavi" { + team_id = (known after apply) + user = "annadavi" } # forgejo_team_repository.cmu-study_study will be created + resource "forgejo_team_repository" "cmu-study_study" { + owner = "ScottyLabs" + repository = "study" + team_id = (known after apply) } # forgejo_team_repository.study_leads_study will be created + resource "forgejo_team_repository" "study_leads_study" { + owner = "ScottyLabs" + repository = "study" + team_id = (known after apply) } # github_actions_repository_permissions.cmu-study_study_actions will be created + resource "github_actions_repository_permissions" "cmu-study_study_actions" { + enabled = false + id = (known after apply) + repository = "study" + sha_pinning_required = (known after apply) } # github_repository.cmu-study_study will be created + resource "github_repository" "cmu-study_study" { + allow_auto_merge = false + allow_forking = (known after apply) + allow_merge_commit = true + allow_rebase_merge = true + allow_squash_merge = true + archived = false + default_branch = (known after apply) + delete_branch_on_merge = false + description = "READ-ONLY MIRROR" + etag = (known after apply) + fork = (known after apply) + full_name = (known after apply) + git_clone_url = (known after apply) + has_discussions = false + has_issues = false + has_projects = false + has_wiki = false + homepage_url = "https://git.cmu.dev/ScottyLabs/study" + html_url = (known after apply) + http_clone_url = (known after apply) + id = (known after apply) + ignore_vulnerability_alerts_during_read = false + merge_commit_message = "PR_TITLE" + merge_commit_title = "MERGE_MESSAGE" + name = "study" + node_id = (known after apply) + primary_language = (known after apply) + private = (known after apply) + repo_id = (known after apply) + source_owner = (known after apply) + source_repo = (known after apply) + squash_merge_commit_message = "COMMIT_MESSAGES" + squash_merge_commit_title = "COMMIT_OR_PR_TITLE" + ssh_clone_url = (known after apply) + svn_url = (known after apply) + topics = (known after apply) + visibility = "public" + vulnerability_alerts = (known after apply) + web_commit_signoff_required = (known after apply) + security_and_analysis (known after apply) } # github_repository_deploy_key.cmu-study_study_mirror_deploy_key will be created + resource "github_repository_deploy_key" "cmu-study_study_mirror_deploy_key" { + etag = (known after apply) + id = (known after apply) + key = (known after apply) + read_only = false + repository = "study" + title = "Forgejo mirror" } # github_repository_ruleset.cmu-study_study_ruleset will be created + resource "github_repository_ruleset" "cmu-study_study_ruleset" { + enforcement = "active" + etag = (known after apply) + id = (known after apply) + name = "Default" + node_id = (known after apply) + repository = "study" + ruleset_id = (known after apply) + target = "branch" + bypass_actors { + actor_id = (known after apply) + actor_type = "DeployKey" + bypass_mode = "always" } + conditions { + ref_name { + exclude = [] + include = [ + "~ALL", ] } } + rules { + creation = true + deletion = true + non_fast_forward = true + update = true + update_allows_fetch_and_merge = false + pull_request { + allowed_merge_methods = (known after apply) + dismiss_stale_reviews_on_push = false + require_code_owner_review = false + require_last_push_approval = false + required_approving_review_count = 1 + required_review_thread_resolution = false } } } # github_team.cmu-study will be created + resource "github_team" "cmu-study" { + create_default_maintainer = false + etag = (known after apply) + id = (known after apply) + members_count = (known after apply) + name = "CMU Study" + node_id = (known after apply) + notification_setting = "notifications_enabled" + parent_team_read_id = (known after apply) + parent_team_read_slug = (known after apply) + privacy = "closed" + slug = (known after apply) } # github_team_membership.cmu-study_annadavi will be created + resource "github_team_membership" "cmu-study_annadavi" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = (known after apply) + username = "annadavid1" } # restapi_object.cmu-study_study_kennel_webhook will be created + resource "restapi_object" "cmu-study_study_kennel_webhook" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = (sensitive value) + id = (known after apply) + id_attribute = "id" + path = "/api/v1/repos/ScottyLabs/study/hooks" + update_method = "PATCH" } # restapi_object.cmu-study_study_push_mirror will be created + resource "restapi_object" "cmu-study_study_push_mirror" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = jsonencode( { + interval = "8h0m0s" + remote_address = "git@github.com:ScottyLabs/study.git" + sync_on_commit = true + use_ssh = true } ) + id = (known after apply) + id_attribute = "remote_name" + path = "/api/v1/repos/ScottyLabs/study/push_mirrors" } Plan: 15 to add, 0 to change, 0 to destroy. ╷ │ Error: Unable to read repository │ │ with forgejo_repository.cmu-eats_dining_api, │ on forgejo_repos.tf.json line 115, in resource.forgejo_repository.cmu-eats_dining_api: │ 115: }, │ │ Repository with ID 65 not found: The target couldn't be found. ╵ ╷ │ Error: Unable to read repository │ │ with forgejo_repository.cmu-eats_eats, │ on forgejo_repos.tf.json line 129, in resource.forgejo_repository.cmu-eats_eats: │ 129: }, │ │ Repository with ID 66 not found: The target couldn't be found. ╵ ``` </details> --- ### Plan Summary 10 projects, 8 with changes, 1 with no changes, 1 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis state -p git rm forgejo_repository.cmu-eats_dining_api forgejo_repository.cmu-eats_eats

atlantis state -p git rm forgejo_repository.cmu-eats_dining_api forgejo_repository.cmu-eats_eats

Ran State rm for project: git dir: tofu/git workspace: default

Removed forgejo_repository.cmu-eats_dining_api
Removed forgejo_repository.cmu-eats_eats
Successfully removed 2 resource instance(s).

🚮 A plan file was discarded. Re-plan would be required before applying.

  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    
Ran State `rm` for project: `git` dir: `tofu/git` workspace: `default` ```diff Removed forgejo_repository.cmu-eats_dining_api Removed forgejo_repository.cmu-eats_eats Successfully removed 2 resource instance(s). ``` :put_litter_in_its_place: A plan file was discarded. Re-plan would be required before applying. * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ```
Owner

atlantis import -p git forgejo_repository.cmu-eats_dining_api ScottyLabs/dining-api

atlantis import -p git forgejo_repository.cmu-eats_dining_api ScottyLabs/dining-api

Ran Import for project: git dir: tofu/git workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.06s

data.external.identity_arom1a: Reading...
data.external.identity_brightz: Reading...
data.external.identity_jc_965: Reading...
data.external.identity_qianxuege: Reading...
data.external.identity_krishsax: Reading...
data.external.identity_emmawong: Reading...
data.external.identity_nikhilr: Reading...
data.external.identity_maybe_yiyi: Reading...
data.external.identity_emmawong: Read complete after 0s [id=-]
data.external.identity_nikhilr: Read complete after 0s [id=-]
data.external.identity_krishsax: Read complete after 0s [id=-]
data.external.identity_gabrielx: Reading...
data.external.identity_dkou: Reading...
data.external.identity_jiyas: Reading...
data.external.identity_qianxuege: Read complete after 0s [id=-]
data.external.identity_siddharthr: Reading...
data.external.identity_brightz: Read complete after 0s [id=-]
data.external.identity_jessica_lr_song: Reading...
data.external.identity_maybe_yiyi: Read complete after 0s [id=-]
data.external.identity_jc_965: Read complete after 0s [id=-]
data.external.identity_manrajs: Reading...
data.external.identity_annadavi: Reading...
data.external.identity_arom1a: Read complete after 0s [id=-]
data.external.identity_rzhu3: Reading...
data.external.identity_gostmeaper: Reading...
data.external.identity_jw8: Reading...
data.external.identity_siddharthr: Read complete after 0s [id=-]
data.external.identity_jalenluorion: Reading...
data.external.identity_gabrielx: Read complete after 0s [id=-]
data.external.identity_dkou: Read complete after 0s [id=-]
data.external.identity_kothurvidur: Reading...
data.external.identity_acchs: Reading...
data.external.identity_annadavi: Read complete after 0s [id=-]
data.external.identity_anish: Reading...
data.external.identity_jiyas: Read complete after 0s [id=-]
data.external.identity_zoexue: Reading...
data.external.identity_rzhu3: Read complete after 0s [id=-]
data.external.identity_kluoxx: Reading...
data.external.identity_manrajs: Read complete after 0s [id=-]
data.external.identity_riztofu: Reading...
data.external.identity_jessica_lr_song: Read complete after 0s [id=-]
data.external.identity_rhsen: Reading...
data.external.identity_gostmeaper: Read complete after 0s [id=-]
data.external.identity_kechezon: Reading...
data.external.identity_jw8: Read complete after 0s [id=-]
data.external.identity_liamwalk: Reading...
data.external.identity_acchs: Read complete after 1s [id=-]
data.external.identity_etashjha: Reading...
data.external.identity_jalenluorion: Read complete after 1s [id=-]
data.external.identity_zoexue: Read complete after 1s [id=-]
data.external.identity_dcrodrig: Reading...
data.external.identity_stringatlas: Reading...
data.external.identity_kothurvidur: Read complete after 1s [id=-]
data.external.identity_xboxbedrock: Reading...
data.external.identity_anish: Read complete after 1s [id=-]
data.external.identity_kritdass: Reading...
data.external.identity_kluoxx: Read complete after 1s [id=-]
data.external.identity_tentype: Reading...
data.external.identity_riztofu: Read complete after 1s [id=-]
data.external.identity_rhsen: Read complete after 1s [id=-]
data.external.identity_thesuperrl: Reading...
data.external.identity_amzoeee: Reading...
data.external.identity_kechezon: Read complete after 1s [id=-]
data.external.identity_liamwalk: Read complete after 1s [id=-]
data.forgejo_organization.this: Reading...
data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs]
data.external.identity_stringatlas: Read complete after 0s [id=-]
data.external.identity_xboxbedrock: Read complete after 0s [id=-]
forgejo_repository.cmu-eats_dining_api: Importing from ID "ScottyLabs/dining-api"...
data.external.identity_dcrodrig: Read complete after 0s [id=-]
data.external.identity_etashjha: Read complete after 0s [id=-]
forgejo_repository.cmu-eats_dining_api: Import prepared!
  Prepared forgejo_repository for import
data.external.identity_kritdass: Read complete after 0s [id=-]
data.external.identity_thesuperrl: Read complete after 0s [id=-]
forgejo_repository.cmu-eats_dining_api: Refreshing state... [name=dining-api]
data.external.identity_amzoeee: Read complete after 0s [id=-]
data.external.identity_tentype: Read complete after 0s [id=-]

Import successful!

The resources that were imported are shown above. These resources are now in
your OpenTofu state and will henceforth be managed by OpenTofu.
🚮 A plan file was discarded. Re-plan would be required before applying.
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    
Ran Import for project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.06s data.external.identity_arom1a: Reading... data.external.identity_brightz: Reading... data.external.identity_jc_965: Reading... data.external.identity_qianxuege: Reading... data.external.identity_krishsax: Reading... data.external.identity_emmawong: Reading... data.external.identity_nikhilr: Reading... data.external.identity_maybe_yiyi: Reading... data.external.identity_emmawong: Read complete after 0s [id=-] data.external.identity_nikhilr: Read complete after 0s [id=-] data.external.identity_krishsax: Read complete after 0s [id=-] data.external.identity_gabrielx: Reading... data.external.identity_dkou: Reading... data.external.identity_jiyas: Reading... data.external.identity_qianxuege: Read complete after 0s [id=-] data.external.identity_siddharthr: Reading... data.external.identity_brightz: Read complete after 0s [id=-] data.external.identity_jessica_lr_song: Reading... data.external.identity_maybe_yiyi: Read complete after 0s [id=-] data.external.identity_jc_965: Read complete after 0s [id=-] data.external.identity_manrajs: Reading... data.external.identity_annadavi: Reading... data.external.identity_arom1a: Read complete after 0s [id=-] data.external.identity_rzhu3: Reading... data.external.identity_gostmeaper: Reading... data.external.identity_jw8: Reading... data.external.identity_siddharthr: Read complete after 0s [id=-] data.external.identity_jalenluorion: Reading... data.external.identity_gabrielx: Read complete after 0s [id=-] data.external.identity_dkou: Read complete after 0s [id=-] data.external.identity_kothurvidur: Reading... data.external.identity_acchs: Reading... data.external.identity_annadavi: Read complete after 0s [id=-] data.external.identity_anish: Reading... data.external.identity_jiyas: Read complete after 0s [id=-] data.external.identity_zoexue: Reading... data.external.identity_rzhu3: Read complete after 0s [id=-] data.external.identity_kluoxx: Reading... data.external.identity_manrajs: Read complete after 0s [id=-] data.external.identity_riztofu: Reading... data.external.identity_jessica_lr_song: Read complete after 0s [id=-] data.external.identity_rhsen: Reading... data.external.identity_gostmeaper: Read complete after 0s [id=-] data.external.identity_kechezon: Reading... data.external.identity_jw8: Read complete after 0s [id=-] data.external.identity_liamwalk: Reading... data.external.identity_acchs: Read complete after 1s [id=-] data.external.identity_etashjha: Reading... data.external.identity_jalenluorion: Read complete after 1s [id=-] data.external.identity_zoexue: Read complete after 1s [id=-] data.external.identity_dcrodrig: Reading... data.external.identity_stringatlas: Reading... data.external.identity_kothurvidur: Read complete after 1s [id=-] data.external.identity_xboxbedrock: Reading... data.external.identity_anish: Read complete after 1s [id=-] data.external.identity_kritdass: Reading... data.external.identity_kluoxx: Read complete after 1s [id=-] data.external.identity_tentype: Reading... data.external.identity_riztofu: Read complete after 1s [id=-] data.external.identity_rhsen: Read complete after 1s [id=-] data.external.identity_thesuperrl: Reading... data.external.identity_amzoeee: Reading... data.external.identity_kechezon: Read complete after 1s [id=-] data.external.identity_liamwalk: Read complete after 1s [id=-] data.forgejo_organization.this: Reading... data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs] data.external.identity_stringatlas: Read complete after 0s [id=-] data.external.identity_xboxbedrock: Read complete after 0s [id=-] forgejo_repository.cmu-eats_dining_api: Importing from ID "ScottyLabs/dining-api"... data.external.identity_dcrodrig: Read complete after 0s [id=-] data.external.identity_etashjha: Read complete after 0s [id=-] forgejo_repository.cmu-eats_dining_api: Import prepared! Prepared forgejo_repository for import data.external.identity_kritdass: Read complete after 0s [id=-] data.external.identity_thesuperrl: Read complete after 0s [id=-] forgejo_repository.cmu-eats_dining_api: Refreshing state... [name=dining-api] data.external.identity_amzoeee: Read complete after 0s [id=-] data.external.identity_tentype: Read complete after 0s [id=-] Import successful! The resources that were imported are shown above. These resources are now in your OpenTofu state and will henceforth be managed by OpenTofu. ``` </details> :put_litter_in_its_place: A plan file was discarded. Re-plan would be required before applying. * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ```
Owner

atlantis import -p git forgejo_repository.cmu-eats_eats ScottyLabs/eats

atlantis import -p git forgejo_repository.cmu-eats_eats ScottyLabs/eats

Ran Import for project: git dir: tofu/git workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.05s

data.external.identity_acchs: Reading...
data.external.identity_zoexue: Reading...
data.external.identity_jalenluorion: Reading...
data.external.identity_jc_965: Reading...
data.external.identity_thesuperrl: Reading...
data.external.identity_liamwalk: Reading...
data.external.identity_emmawong: Reading...
data.external.identity_siddharthr: Reading...
data.external.identity_jc_965: Read complete after 0s [id=-]
data.external.identity_arom1a: Reading...
data.external.identity_zoexue: Read complete after 0s [id=-]
data.external.identity_amzoeee: Reading...
data.external.identity_acchs: Read complete after 0s [id=-]
data.external.identity_xboxbedrock: Reading...
data.external.identity_liamwalk: Read complete after 0s [id=-]
data.external.identity_manrajs: Reading...
data.external.identity_thesuperrl: Read complete after 0s [id=-]
data.external.identity_jessica_lr_song: Reading...
data.external.identity_emmawong: Read complete after 0s [id=-]
data.external.identity_rzhu3: Reading...
data.external.identity_siddharthr: Read complete after 0s [id=-]
data.external.identity_kothurvidur: Reading...
data.external.identity_jalenluorion: Read complete after 0s [id=-]
data.external.identity_stringatlas: Reading...
data.external.identity_dkou: Reading...
data.external.identity_krishsax: Reading...
data.external.identity_arom1a: Read complete after 0s [id=-]
data.external.identity_amzoeee: Read complete after 0s [id=-]
data.external.identity_kluoxx: Reading...
data.external.identity_maybe_yiyi: Reading...
data.external.identity_xboxbedrock: Read complete after 0s [id=-]
data.external.identity_rhsen: Reading...
data.external.identity_jessica_lr_song: Read complete after 1s [id=-]
data.external.identity_dcrodrig: Reading...
data.external.identity_kothurvidur: Read complete after 1s [id=-]
data.external.identity_jw8: Reading...
data.external.identity_stringatlas: Read complete after 1s [id=-]
data.external.identity_rzhu3: Read complete after 1s [id=-]
data.external.identity_annadavi: Reading...
data.external.identity_gabrielx: Reading...
data.external.identity_manrajs: Read complete after 1s [id=-]
data.external.identity_nikhilr: Reading...
data.external.identity_krishsax: Read complete after 1s [id=-]
data.external.identity_dkou: Read complete after 1s [id=-]
data.external.identity_brightz: Reading...
data.external.identity_jiyas: Reading...
data.external.identity_maybe_yiyi: Read complete after 1s [id=-]
data.external.identity_tentype: Reading...
data.external.identity_kluoxx: Read complete after 1s [id=-]
data.external.identity_anish: Reading...
data.external.identity_rhsen: Read complete after 1s [id=-]
data.external.identity_gostmeaper: Reading...
data.external.identity_dcrodrig: Read complete after 0s [id=-]
data.external.identity_riztofu: Reading...
data.external.identity_gabrielx: Read complete after 0s [id=-]
data.external.identity_kritdass: Reading...
data.external.identity_jw8: Read complete after 0s [id=-]
data.external.identity_etashjha: Reading...
data.external.identity_annadavi: Read complete after 0s [id=-]
data.external.identity_qianxuege: Reading...
data.external.identity_nikhilr: Read complete after 0s [id=-]
data.external.identity_kechezon: Reading...
data.external.identity_jiyas: Read complete after 0s [id=-]
forgejo_repository.cmu-eats_eats: Importing from ID "ScottyLabs/eats"...
data.external.identity_anish: Read complete after 0s [id=-]
forgejo_repository.cmu-eats_eats: Import prepared!
  Prepared forgejo_repository for import
data.external.identity_tentype: Read complete after 0s [id=-]
data.external.identity_gostmeaper: Read complete after 0s [id=-]
data.forgejo_organization.this: Reading...
data.external.identity_riztofu: Read complete after 0s [id=-]
data.external.identity_brightz: Read complete after 0s [id=-]
data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs]
forgejo_repository.cmu-eats_eats: Refreshing state... [name=eats]
data.external.identity_etashjha: Read complete after 0s [id=-]
data.external.identity_qianxuege: Read complete after 0s [id=-]
data.external.identity_kechezon: Read complete after 0s [id=-]
data.external.identity_kritdass: Read complete after 0s [id=-]

Import successful!

The resources that were imported are shown above. These resources are now in
your OpenTofu state and will henceforth be managed by OpenTofu.
🚮 A plan file was discarded. Re-plan would be required before applying.
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    
Ran Import for project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.05s data.external.identity_acchs: Reading... data.external.identity_zoexue: Reading... data.external.identity_jalenluorion: Reading... data.external.identity_jc_965: Reading... data.external.identity_thesuperrl: Reading... data.external.identity_liamwalk: Reading... data.external.identity_emmawong: Reading... data.external.identity_siddharthr: Reading... data.external.identity_jc_965: Read complete after 0s [id=-] data.external.identity_arom1a: Reading... data.external.identity_zoexue: Read complete after 0s [id=-] data.external.identity_amzoeee: Reading... data.external.identity_acchs: Read complete after 0s [id=-] data.external.identity_xboxbedrock: Reading... data.external.identity_liamwalk: Read complete after 0s [id=-] data.external.identity_manrajs: Reading... data.external.identity_thesuperrl: Read complete after 0s [id=-] data.external.identity_jessica_lr_song: Reading... data.external.identity_emmawong: Read complete after 0s [id=-] data.external.identity_rzhu3: Reading... data.external.identity_siddharthr: Read complete after 0s [id=-] data.external.identity_kothurvidur: Reading... data.external.identity_jalenluorion: Read complete after 0s [id=-] data.external.identity_stringatlas: Reading... data.external.identity_dkou: Reading... data.external.identity_krishsax: Reading... data.external.identity_arom1a: Read complete after 0s [id=-] data.external.identity_amzoeee: Read complete after 0s [id=-] data.external.identity_kluoxx: Reading... data.external.identity_maybe_yiyi: Reading... data.external.identity_xboxbedrock: Read complete after 0s [id=-] data.external.identity_rhsen: Reading... data.external.identity_jessica_lr_song: Read complete after 1s [id=-] data.external.identity_dcrodrig: Reading... data.external.identity_kothurvidur: Read complete after 1s [id=-] data.external.identity_jw8: Reading... data.external.identity_stringatlas: Read complete after 1s [id=-] data.external.identity_rzhu3: Read complete after 1s [id=-] data.external.identity_annadavi: Reading... data.external.identity_gabrielx: Reading... data.external.identity_manrajs: Read complete after 1s [id=-] data.external.identity_nikhilr: Reading... data.external.identity_krishsax: Read complete after 1s [id=-] data.external.identity_dkou: Read complete after 1s [id=-] data.external.identity_brightz: Reading... data.external.identity_jiyas: Reading... data.external.identity_maybe_yiyi: Read complete after 1s [id=-] data.external.identity_tentype: Reading... data.external.identity_kluoxx: Read complete after 1s [id=-] data.external.identity_anish: Reading... data.external.identity_rhsen: Read complete after 1s [id=-] data.external.identity_gostmeaper: Reading... data.external.identity_dcrodrig: Read complete after 0s [id=-] data.external.identity_riztofu: Reading... data.external.identity_gabrielx: Read complete after 0s [id=-] data.external.identity_kritdass: Reading... data.external.identity_jw8: Read complete after 0s [id=-] data.external.identity_etashjha: Reading... data.external.identity_annadavi: Read complete after 0s [id=-] data.external.identity_qianxuege: Reading... data.external.identity_nikhilr: Read complete after 0s [id=-] data.external.identity_kechezon: Reading... data.external.identity_jiyas: Read complete after 0s [id=-] forgejo_repository.cmu-eats_eats: Importing from ID "ScottyLabs/eats"... data.external.identity_anish: Read complete after 0s [id=-] forgejo_repository.cmu-eats_eats: Import prepared! Prepared forgejo_repository for import data.external.identity_tentype: Read complete after 0s [id=-] data.external.identity_gostmeaper: Read complete after 0s [id=-] data.forgejo_organization.this: Reading... data.external.identity_riztofu: Read complete after 0s [id=-] data.external.identity_brightz: Read complete after 0s [id=-] data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs] forgejo_repository.cmu-eats_eats: Refreshing state... [name=eats] data.external.identity_etashjha: Read complete after 0s [id=-] data.external.identity_qianxuege: Read complete after 0s [id=-] data.external.identity_kechezon: Read complete after 0s [id=-] data.external.identity_kritdass: Read complete after 0s [id=-] Import successful! The resources that were imported are shown above. These resources are now in your OpenTofu state and will henceforth be managed by OpenTofu. ``` </details> :put_litter_in_its_place: A plan file was discarded. Re-plan would be required before applying. * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ```
Owner

atlantis plan -p git

atlantis plan -p git

Ran Plan for dir: tofu/git workspace: default

Plan Error

Show Output
running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu plan -input=false -refresh -out /var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git/git-default.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git': exit status 1
    Finished `release` profile [optimized] target(s) in 0.05s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
random_password.kennel_webhook_secret: Refreshing state... [id=none]
forgejo_repository.cmu-courses_courses: Refreshing state... [name=courses]
forgejo_repository.devops_governance: Refreshing state... [name=governance]
forgejo_repository.cmu-eats_dining_api: Refreshing state... [name=dining-api]
forgejo_repository.cmu-maps_maps: Refreshing state... [name=maps]
forgejo_repository.cmu-courses_internet_archive: Refreshing state... [name=internet-archive]
forgejo_repository.quest_quest: Refreshing state... [name=quest]
forgejo_repository.uia_scottylabs_org: Refreshing state... [name=scottylabs-org]
forgejo_repository.devops_infrastructure: Refreshing state... [name=infrastructure]
forgejo_repository.cmu-housing_housing: Refreshing state... [name=housing]
forgejo_repository.slai_cmugpt_surface: Refreshing state... [name=cmugpt-surface]
forgejo_repository.terrier_terrier: Refreshing state... [name=terrier]
forgejo_repository.cbp_groupme_mirror: Refreshing state... [name=groupme-mirror]
data.forgejo_organization.this: Reading...
forgejo_repository.devops_kennel: Refreshing state... [name=kennel]
forgejo_repository.cmu-cal_cal: Refreshing state... [name=cal]
forgejo_repository.tartan-vote_tartan_vote: Refreshing state... [name=tartan-vote]
data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs]
forgejo_repository.cbp_discord_verify: Refreshing state... [name=discord-verify]
forgejo_repository.lost-and-found_lost_and_found_v2: Refreshing state... [name=lost-and-found-v2]
forgejo_repository.slai_cmugpt_sms_surface: Refreshing state... [name=cmugpt-sms-surface]
forgejo_repository.slai_mcp_server: Refreshing state... [name=mcp-server]
forgejo_repository.slai_cmugpt_agent: Refreshing state... [name=cmugpt-agent]
forgejo_repository.occ_occ: Refreshing state... [name=occ]
forgejo_repository.cbp_dalmatian: Refreshing state... [name=dalmatian]
forgejo_repository.devops_documentation: Refreshing state... [name=documentation]
forgejo_repository.devops_collie: Refreshing state... [name=collie]
forgejo_repository.devops_link_shortener: Refreshing state... [name=link-shortener]
forgejo_repository.uia_components: Refreshing state... [name=components]
forgejo_repository.cbp_bus_sign: Refreshing state... [name=bus-sign]
forgejo_repository.cmu-eats_eats: Refreshing state... [name=eats]
forgejo_repository.cmu-insta_insta: Refreshing state... [name=insta]
data.external.identity_riztofu: Reading...
data.external.identity_kechezon: Reading...
data.external.identity_dcrodrig: Reading...
data.external.identity_acchs: Reading...
data.external.identity_gostmeaper: Reading...
data.external.identity_liamwalk: Reading...
data.external.identity_etashjha: Reading...
data.external.identity_kluoxx: Reading...
data.external.identity_xboxbedrock: Reading...
data.external.identity_gabrielx: Reading...
data.external.identity_riztofu: Read complete after 0s [id=-]
data.external.identity_jw8: Reading...
data.external.identity_acchs: Read complete after 0s [id=-]
data.external.identity_tentype: Reading...
data.external.identity_kechezon: Read complete after 0s [id=-]
data.external.identity_brightz: Reading...
data.external.identity_dcrodrig: Read complete after 0s [id=-]
data.external.identity_jalenluorion: Reading...
data.external.identity_etashjha: Read complete after 0s [id=-]
data.external.identity_jessica_lr_song: Reading...
data.external.identity_liamwalk: Read complete after 0s [id=-]
data.external.identity_gabrielx: Read complete after 0s [id=-]
data.external.identity_amzoeee: Reading...
data.external.identity_zoexue: Reading...
data.external.identity_xboxbedrock: Read complete after 0s [id=-]
data.external.identity_jc_965: Reading...
data.external.identity_kluoxx: Read complete after 0s [id=-]
data.external.identity_anish: Reading...
data.external.identity_gostmeaper: Read complete after 0s [id=-]
data.external.identity_arom1a: Reading...
data.external.identity_jw8: Read complete after 0s [id=-]
data.external.identity_jiyas: Reading...
data.external.identity_tentype: Read complete after 0s [id=-]
data.external.identity_krishsax: Reading...
data.external.identity_jessica_lr_song: Read complete after 0s [id=-]
data.external.identity_jalenluorion: Read complete after 0s [id=-]
data.external.identity_annadavi: Reading...
data.external.identity_qianxuege: Reading...
data.external.identity_brightz: Read complete after 0s [id=-]
data.external.identity_maybe_yiyi: Reading...
data.external.identity_amzoeee: Read complete after 0s [id=-]
data.external.identity_kritdass: Reading...
data.external.identity_zoexue: Read complete after 0s [id=-]
data.external.identity_dkou: Reading...
data.external.identity_jc_965: Read complete after 0s [id=-]
data.external.identity_siddharthr: Reading...
data.external.identity_anish: Read complete after 0s [id=-]
data.external.identity_manrajs: Reading...
data.external.identity_arom1a: Read complete after 0s [id=-]
data.external.identity_kothurvidur: Reading...
data.external.identity_jiyas: Read complete after 1s [id=-]
data.external.identity_emmawong: Reading...
data.external.identity_maybe_yiyi: Read complete after 1s [id=-]
data.external.identity_thesuperrl: Reading...
data.external.identity_krishsax: Read complete after 1s [id=-]
data.external.identity_annadavi: Read complete after 1s [id=-]
data.external.identity_stringatlas: Reading...
data.external.identity_rzhu3: Reading...
data.external.identity_qianxuege: Read complete after 1s [id=-]
data.external.identity_nikhilr: Reading...
data.external.identity_dkou: Read complete after 1s [id=-]
data.external.identity_rhsen: Reading...
data.external.identity_kritdass: Read complete after 1s [id=-]
restapi_object.cmu-eats_dining_api_push_mirror: Refreshing state... [id=remote_mirror_a4MsQ5rO1_Z]
data.external.identity_siddharthr: Read complete after 1s [id=-]
restapi_object.devops_governance_kennel_webhook: Refreshing state... [id=22]
data.external.identity_manrajs: Read complete after 1s [id=-]
restapi_object.devops_governance_push_mirror: Refreshing state... [id=remote_mirror_lt9JzkX1jMq]
data.external.identity_kothurvidur: Read complete after 1s [id=-]
restapi_object.devops_infrastructure_push_mirror: Refreshing state... [id=remote_mirror_oYtFpKVzTaZ]
restapi_object.devops_infrastructure_kennel_webhook: Refreshing state... [id=26]
restapi_object.cmu-courses_courses_push_mirror: Refreshing state... [id=remote_mirror_WMR9yQwKFDF]
restapi_object.cmu-courses_courses_kennel_webhook: Refreshing state... [id=11]
restapi_object.cmu-maps_maps_push_mirror: Refreshing state... [id=remote_mirror_e4_FUnvoslJ]
restapi_object.cmu-maps_maps_kennel_webhook: Refreshing state... [id=9]
restapi_object.uia_scottylabs_org_kennel_webhook: Refreshing state... [id=29]
restapi_object.cmu-courses_internet_archive_push_mirror: Refreshing state... [id=remote_mirror_58X__Yfs0bR]
restapi_object.uia_scottylabs_org_push_mirror: Refreshing state... [id=remote_mirror_h6yN6QdS1lD]
restapi_object.quest_quest_kennel_webhook: Refreshing state... [id=10]
restapi_object.quest_quest_push_mirror: Refreshing state... [id=remote_mirror_U9d9BPAon29]
data.external.identity_emmawong: Read complete after 0s [id=-]
forgejo_team.terrier: Refreshing state... [name=terrier]
forgejo_team.cmugpt-sms-surface_leads: Refreshing state... [name=cmugpt-sms-surface-leads]
forgejo_team.cmu-eats: Refreshing state... [name=cmu-eats]
forgejo_team.bus-sign_leads: Refreshing state... [name=bus-sign-leads]
forgejo_team.cmu-housing: Refreshing state... [name=cmu-housing]
data.external.identity_stringatlas: Read complete after 0s [id=-]
forgejo_team.cmu-courses_leads: Refreshing state... [name=cmu-courses-leads]
forgejo_team.documentation_leads: Refreshing state... [name=documentation-leads]
data.external.identity_rhsen: Read complete after 0s [id=-]
data.external.identity_nikhilr: Read complete after 0s [id=-]
data.external.identity_rzhu3: Read complete after 0s [id=-]
forgejo_team.devops_leads: Refreshing state... [name=devops-leads]
forgejo_team.uia: Refreshing state... [name=uia]
forgejo_team.scottylabs-org_leads: Refreshing state... [name=scottylabs-org-leads]
forgejo_team.cmu-cal: Refreshing state... [name=cmu-cal]
forgejo_team.eats_leads: Refreshing state... [name=eats-leads]
forgejo_team.quest_leads: Refreshing state... [name=quest-leads]
forgejo_team.cmu-courses: Refreshing state... [name=cmu-courses]
forgejo_team.occ: Refreshing state... [name=occ]
data.external.identity_thesuperrl: Read complete after 0s [id=-]
forgejo_team.occ-main_leads: Refreshing state... [name=occ-main-leads]
forgejo_team.quest: Refreshing state... [name=quest]
forgejo_team.lost-and-found-v2_leads: Refreshing state... [name=lost-and-found-v2-leads]
forgejo_team.slai: Refreshing state... [name=slai]
forgejo_team.tartan-vote_leads: Refreshing state... [name=tartan-vote-leads]
forgejo_team.courses_leads: Refreshing state... [name=courses-leads]
forgejo_team.devops: Refreshing state... [name=devops]
forgejo_team.cmugpt-agent_leads: Refreshing state... [name=cmugpt-agent-leads]
forgejo_team.cmu-insta: Refreshing state... [name=cmu-insta]
forgejo_team.maps_leads: Refreshing state... [name=maps-leads]
forgejo_team.dalmatian_leads: Refreshing state... [name=dalmatian-leads]
forgejo_team.insta_leads: Refreshing state... [name=insta-leads]
forgejo_team.tartan-vote: Refreshing state... [name=tartan-vote]
forgejo_team.terrier_leads: Refreshing state... [name=terrier-leads]
forgejo_team.cbp: Refreshing state... [name=cbp]
forgejo_team.discord-verify_leads: Refreshing state... [name=discord-verify-leads]
forgejo_team.cmu-maps: Refreshing state... [name=cmu-maps]
forgejo_team.components_leads: Refreshing state... [name=components-leads]
forgejo_team.link-shortener_leads: Refreshing state... [name=link-shortener-leads]
forgejo_team.cal_leads: Refreshing state... [name=cal-leads]
forgejo_team.lost-and-found: Refreshing state... [name=lost-and-found]
forgejo_team.cmugpt-surface_leads: Refreshing state... [name=cmugpt-surface-leads]
forgejo_team.housing_leads: Refreshing state... [name=housing-leads]
forgejo_team.cmugpt-mcp-server_leads: Refreshing state... [name=cmugpt-mcp-server-leads]
forgejo_team.groupme-mirror_leads: Refreshing state... [name=groupme-mirror-leads]
restapi_object.terrier_terrier_push_mirror: Refreshing state... [id=remote_mirror_YopDgZksZv5]
restapi_object.terrier_terrier_kennel_webhook: Refreshing state... [id=12]
restapi_object.slai_cmugpt_surface_kennel_webhook: Refreshing state... [id=23]
restapi_object.slai_cmugpt_surface_push_mirror: Refreshing state... [id=remote_mirror_cPTQN6S5SpB]
restapi_object.cmu-housing_housing_push_mirror: Refreshing state... [id=remote_mirror_uoV2HG4aiOT]
restapi_object.cmu-housing_housing_kennel_webhook: Refreshing state... [id=16]
restapi_object.cmu-housing_housing_docs_webhook: Refreshing state... [id=15]
restapi_object.tartan-vote_tartan_vote_push_mirror: Refreshing state... [id=remote_mirror_EUFq4sFSnH5]
restapi_object.tartan-vote_tartan_vote_kennel_webhook: Refreshing state... [id=24]
restapi_object.tartan-vote_tartan_vote_docs_webhook: Refreshing state... [id=25]
restapi_object.devops_kennel_push_mirror: Refreshing state... [id=remote_mirror_cT9psJkhTQ4]
restapi_object.devops_kennel_kennel_webhook: Refreshing state... [id=14]
restapi_object.cbp_groupme_mirror_push_mirror: Refreshing state... [id=remote_mirror_U61HJHB9VGE]
restapi_object.lost-and-found_lost_and_found_v2_push_mirror: Refreshing state... [id=remote_mirror_n_UZBFU-CY4]
restapi_object.cmu-cal_cal_kennel_webhook: Refreshing state... [id=20]
restapi_object.cmu-cal_cal_push_mirror: Refreshing state... [id=remote_mirror_CO243uIONoq]
restapi_object.cbp_discord_verify_kennel_webhook: Refreshing state... [id=18]
restapi_object.cbp_discord_verify_push_mirror: Refreshing state... [id=remote_mirror_3MV7Up6cY1X]
restapi_object.slai_cmugpt_sms_surface_push_mirror: Refreshing state... [id=remote_mirror__fkBkASN2jA]
restapi_object.slai_cmugpt_sms_surface_kennel_webhook: Refreshing state... [id=19]
restapi_object.slai_mcp_server_push_mirror: Refreshing state... [id=remote_mirror_Tqh2y0R_vDp]
restapi_object.slai_mcp_server_kennel_webhook: Refreshing state... [id=13]
restapi_object.slai_cmugpt_agent_push_mirror: Refreshing state... [id=remote_mirror_kNY6rVcxdyr]
restapi_object.slai_cmugpt_agent_kennel_webhook: Refreshing state... [id=28]
github_team.cbp: Refreshing state... [id=17170099]
github_repository.cmu-eats_eats: Refreshing state... [id=eats]
github_repository.devops_kennel: Refreshing state... [id=kennel]
github_repository.devops_collie: Refreshing state... [id=collie]
github_repository.cmu-insta_insta: Refreshing state... [id=insta]
github_repository.cmu-eats_dining_api: Refreshing state... [id=dining-api]
github_team.cmu-courses: Refreshing state... [id=3643907]
github_team.occ: Refreshing state... [id=19240310]
github_team.cmu-eats: Refreshing state... [id=16807600]
github_team.cmu-insta: Refreshing state... [id=19240311]
github_repository.cmu-cal_cal: Refreshing state... [id=cal]
github_repository.devops_link_shortener: Refreshing state... [id=link-shortener]
github_repository.occ_occ: Refreshing state... [id=occ]
github_repository.devops_infrastructure: Refreshing state... [id=infrastructure]
github_repository.slai_cmugpt_surface: Refreshing state... [id=cmugpt-surface]
github_repository.cbp_groupme_mirror: Refreshing state... [id=groupme-mirror]
github_repository.slai_cmugpt_sms_surface: Refreshing state... [id=cmugpt-sms-surface]
github_repository.cbp_dalmatian: Refreshing state... [id=dalmatian]
github_team.terrier: Refreshing state... [id=12028577]
github_team.devops: Refreshing state... [id=14706218]
github_repository.uia_components: Refreshing state... [id=components]
github_repository.lost-and-found_lost_and_found_v2: Refreshing state... [id=lost-and-found-v2]
github_repository.cbp_discord_verify: Refreshing state... [id=discord-verify]
github_team.cmu-cal: Refreshing state... [id=15669112]
github_repository.quest_quest: Refreshing state... [id=quest]
github_repository.uia_scottylabs_org: Refreshing state... [id=scottylabs-org]
github_repository.cmu-housing_housing: Refreshing state... [id=housing]
github_team.uia: Refreshing state... [id=17529763]
github_repository.terrier_terrier: Refreshing state... [id=terrier]
github_repository.devops_governance: Refreshing state... [id=governance]
github_team.slai: Refreshing state... [id=17571328]
github_repository.slai_cmugpt_agent: Refreshing state... [id=cmugpt-agent]
github_repository.devops_documentation: Refreshing state... [id=documentation]
github_repository.tartan-vote_tartan_vote: Refreshing state... [id=tartan-vote]
github_repository.cmu-courses_courses: Refreshing state... [id=courses]
github_team.lost-and-found: Refreshing state... [id=4088608]
github_team.cmu-maps: Refreshing state... [id=10872360]
github_repository.cbp_bus_sign: Refreshing state... [id=bus-sign]
github_repository.slai_mcp_server: Refreshing state... [id=mcp-server]
github_repository.cmu-courses_internet_archive: Refreshing state... [id=internet-archive]
github_repository.cmu-maps_maps: Refreshing state... [id=maps]
github_team.tartan-vote: Refreshing state... [id=17701833]
github_team.cmu-housing: Refreshing state... [id=17937919]
github_team.quest: Refreshing state... [id=17330197]
restapi_object.devops_documentation_push_mirror: Refreshing state... [id=remote_mirror_izmsqaNB6Wx]
restapi_object.occ_occ_push_mirror: Refreshing state... [id=remote_mirror_moBLq1YG1qz]
restapi_object.cbp_dalmatian_kennel_webhook: Refreshing state... [id=6]
restapi_object.cbp_dalmatian_push_mirror: Refreshing state... [id=remote_mirror_qzlCpLK153F]
restapi_object.cbp_dalmatian_docs_webhook: Refreshing state... [id=7]
restapi_object.devops_link_shortener_push_mirror: Refreshing state... [id=remote_mirror_cp8daMpt82v]
restapi_object.devops_link_shortener_kennel_webhook: Refreshing state... [id=8]
restapi_object.devops_collie_push_mirror: Refreshing state... [id=remote_mirror_ljvVejNFXaS]
restapi_object.devops_collie_kennel_webhook: Refreshing state... [id=17]
restapi_object.uia_components_push_mirror: Refreshing state... [id=remote_mirror_Kwwa8TDnTgc]
restapi_object.uia_components_kennel_webhook: Refreshing state... [id=27]
restapi_object.cmu-eats_eats_kennel_webhook: Refreshing state... [id=31]
restapi_object.cmu-eats_eats_push_mirror: Refreshing state... [id=remote_mirror_sjYwMA6NPgh]
restapi_object.cmu-insta_insta_push_mirror: Refreshing state... [id=remote_mirror_891nu6XQ1us]
restapi_object.cbp_bus_sign_kennel_webhook: Refreshing state... [id=21]
restapi_object.cbp_bus_sign_push_mirror: Refreshing state... [id=remote_mirror_MkKxQ2vA9pc]
forgejo_team_member.cmugpt-sms-surface_leads_krishsax: Refreshing state...
forgejo_team_member.cmugpt-sms-surface_leads_anish: Refreshing state...
forgejo_team_member.terrier_amzoeee: Refreshing state...
forgejo_team_member.terrier_dkou: Refreshing state...
forgejo_team_member.terrier_anish: Refreshing state...
forgejo_team_member.terrier_etashjha: Refreshing state...
forgejo_team_member.terrier_kritdass: Refreshing state...
forgejo_team_member.terrier_kothurvidur: Refreshing state...
forgejo_team_repository.terrier_terrier: Refreshing state...
forgejo_team_member.cmu-housing_rhsen: Refreshing state...
forgejo_team_member.cmu-housing_kluoxx: Refreshing state...
forgejo_team_member.cmu-housing_nikhilr: Refreshing state...
forgejo_team_member.cmu-housing_stringatlas: Refreshing state...
forgejo_team_member.cmu-housing_manrajs: Refreshing state...
forgejo_team_member.cmu-housing_amzoeee: Refreshing state...
forgejo_team_member.cmu-housing_jessica_lr_song: Refreshing state...
forgejo_team_member.cmu-housing_anish: Refreshing state...
forgejo_team_member.cmu-housing_tentype: Refreshing state...
forgejo_team_member.bus-sign_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.bus-sign_leads_anish: Refreshing state...
forgejo_team_member.bus-sign_leads_tentype: Refreshing state...
forgejo_team_member.cmu-housing_emmawong: Refreshing state...
forgejo_team_member.cmu-housing_gostmeaper: Refreshing state...
forgejo_team_member.cmu-housing_zoexue: Refreshing state...
forgejo_team_repository.cmu-housing_housing: Refreshing state...
forgejo_team_repository.cmu-eats_dining_api: Refreshing state...
forgejo_team_member.cmu-housing_xboxbedrock: Refreshing state...
forgejo_team_repository.cmu-eats_eats: Refreshing state...
forgejo_team_member.cmu-eats_arom1a: Refreshing state...
forgejo_team_member.cmu-courses_leads_etashjha: Refreshing state...
forgejo_team_member.devops_leads_anish: Refreshing state...
forgejo_team_member.cmu-courses_leads_anish: Refreshing state...
forgejo_team_member.documentation_leads_anish: Refreshing state...
forgejo_team_repository.uia_scottylabs_org: Refreshing state...
forgejo_team_member.uia_anish: Refreshing state...
forgejo_team_member.documentation_leads_thesuperrl: Refreshing state...
forgejo_team_member.devops_leads_thesuperrl: Refreshing state...
forgejo_team_repository.uia_components: Refreshing state...
forgejo_team_member.scottylabs-org_leads_anish: Refreshing state...
forgejo_team_member.cmu-cal_anish: Refreshing state...
forgejo_team_member.cmu-cal_jalenluorion: Refreshing state...
forgejo_team_repository.cmu-cal_cal: Refreshing state...
forgejo_team_member.occ_dcrodrig: Refreshing state...
forgejo_team_member.occ_anish: Refreshing state...
forgejo_team_member.occ_xboxbedrock: Refreshing state...
forgejo_team_repository.occ_occ: Refreshing state...
forgejo_team_member.occ_acchs: Refreshing state...
forgejo_team_member.lost-and-found-v2_leads_anish: Refreshing state...
forgejo_team_member.quest_xboxbedrock: Refreshing state...
forgejo_team_member.quest_kechezon: Refreshing state...
forgejo_team_member.quest_jw8: Refreshing state...
forgejo_team_member.quest_brightz: Refreshing state...
forgejo_team_repository.quest_quest: Refreshing state...
forgejo_team_member.quest_anish: Refreshing state...
forgejo_team_member.slai_jc_965: Refreshing state...
forgejo_team_repository.slai_cmugpt_surface: Refreshing state...
forgejo_team_member.slai_anish: Refreshing state...
forgejo_team_member.slai_qianxuege: Refreshing state...
forgejo_team_repository.slai_cmugpt_sms_surface: Refreshing state...
forgejo_team_repository.slai_cmugpt_agent: Refreshing state...
forgejo_team_member.slai_krishsax: Refreshing state...
forgejo_team_repository.slai_mcp_server: Refreshing state...
forgejo_team_member.slai_rzhu3: Refreshing state...
forgejo_team_member.slai_xboxbedrock: Refreshing state...
forgejo_team_member.slai_jiyas: Refreshing state...
forgejo_team_member.cmu-courses_gabrielx: Refreshing state...
forgejo_team_member.cmu-courses_dkou: Refreshing state...
forgejo_team_member.cmu-courses_etashjha: Refreshing state...
forgejo_team_repository.cmu-courses_internet_archive: Refreshing state...
forgejo_team_member.cmu-courses_anish: Refreshing state...
forgejo_team_repository.cmu-courses_courses: Refreshing state...
forgejo_team_member.eats_leads_arom1a: Refreshing state...
forgejo_team_member.occ-main_leads_anish: Refreshing state...
forgejo_team_member.occ-main_leads_dcrodrig: Refreshing state...
forgejo_team_member.courses_leads_anish: Refreshing state...
forgejo_team_member.courses_leads_etashjha: Refreshing state...
forgejo_team_repository.devops_infrastructure: Refreshing state...
forgejo_team_repository.devops_governance: Refreshing state...
forgejo_team_repository.devops_link_shortener: Refreshing state...
forgejo_team_member.devops_anish: Refreshing state...
forgejo_team_repository.devops_collie: Refreshing state...
forgejo_team_repository.devops_kennel: Refreshing state...
forgejo_team_member.devops_xboxbedrock: Refreshing state...
forgejo_team_repository.devops_documentation: Refreshing state...
forgejo_team_member.devops_thesuperrl: Refreshing state...
forgejo_team_member.tartan-vote_leads_anish: Refreshing state...
forgejo_team_member.quest_leads_xboxbedrock: Refreshing state...
forgejo_team_member.tartan-vote_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.quest_leads_anish: Refreshing state...
forgejo_team_repository.cmu-insta_insta: Refreshing state...
forgejo_team_member.cmu-insta_etashjha: Refreshing state...
forgejo_team_member.maps_leads_anish: Refreshing state...
forgejo_team_member.maps_leads_siddharthr: Refreshing state...
forgejo_team_member.cmugpt-agent_leads_krishsax: Refreshing state...
forgejo_team_member.cmugpt-agent_leads_anish: Refreshing state...
forgejo_team_member.dalmatian_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.dalmatian_leads_tentype: Refreshing state...
forgejo_team_member.dalmatian_leads_anish: Refreshing state...
forgejo_team_member.insta_leads_etashjha: Refreshing state...
forgejo_team_member.discord-verify_leads_tentype: Refreshing state...
forgejo_team_member.discord-verify_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.discord-verify_leads_anish: Refreshing state...
forgejo_team_member.cbp_tentype: Refreshing state...
forgejo_team_member.cbp_xboxbedrock: Refreshing state...
forgejo_team_member.cbp_nikhilr: Refreshing state...
forgejo_team_member.cbp_maybe_yiyi: Refreshing state...
forgejo_team_member.cbp_liamwalk: Refreshing state...
forgejo_team_repository.cbp_discord_verify: Refreshing state...
forgejo_team_repository.cbp_bus_sign: Refreshing state...
forgejo_team_member.cbp_anish: Refreshing state...
forgejo_team_repository.cbp_dalmatian: Refreshing state...
forgejo_team_repository.cbp_groupme_mirror: Refreshing state...
forgejo_team_member.tartan-vote_maybe_yiyi: Refreshing state...
forgejo_team_member.tartan-vote_xboxbedrock: Refreshing state...
forgejo_team_member.tartan-vote_riztofu: Refreshing state...
forgejo_team_repository.tartan-vote_tartan_vote: Refreshing state...
forgejo_team_member.tartan-vote_anish: Refreshing state...
forgejo_team_member.tartan-vote_tentype: Refreshing state...
forgejo_team_member.cmu-maps_anish: Refreshing state...
forgejo_team_repository.cmu-maps_maps: Refreshing state...
forgejo_team_member.cmu-maps_siddharthr: Refreshing state...
forgejo_team_member.cal_leads_jalenluorion: Refreshing state...
forgejo_team_member.cal_leads_anish: Refreshing state...
forgejo_team_member.terrier_leads_anish: Refreshing state...
forgejo_team_member.terrier_leads_kritdass: Refreshing state...
forgejo_team_member.cmugpt-surface_leads_krishsax: Refreshing state...
forgejo_team_member.cmugpt-surface_leads_anish: Refreshing state...
forgejo_team_member.components_leads_anish: Refreshing state...
forgejo_team_repository.lost-and-found_lost_and_found_v2: Refreshing state...
forgejo_team_member.lost-and-found_anish: Refreshing state...
forgejo_team_member.housing_leads_emmawong: Refreshing state...
forgejo_team_member.housing_leads_anish: Refreshing state...
forgejo_team_member.housing_leads_tentype: Refreshing state...
forgejo_team_member.housing_leads_nikhilr: Refreshing state...
forgejo_team_member.housing_leads_gostmeaper: Refreshing state...
forgejo_team_member.link-shortener_leads_anish: Refreshing state...
forgejo_team_member.link-shortener_leads_thesuperrl: Refreshing state...
forgejo_team_member.groupme-mirror_leads_anish: Refreshing state...
forgejo_team_member.groupme-mirror_leads_tentype: Refreshing state...
forgejo_team_member.groupme-mirror_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.cmugpt-mcp-server_leads_anish: Refreshing state...
forgejo_team_member.cmugpt-mcp-server_leads_krishsax: Refreshing state...
github_team_membership.cbp_liamwalk: Refreshing state... [id=17170099:pandaaured]
github_team_membership.cbp_anish: Refreshing state... [id=17170099:ap-1]
github_team_membership.cbp_maybe_yiyi: Refreshing state... [id=17170099:maybe-yiyi]
github_team_membership.cbp_xboxbedrock: Refreshing state... [id=17170099:xboxbedrock]
github_team_membership.cbp_tentype: Refreshing state... [id=17170099:tentype]
github_team_membership.cbp_nikhilr: Refreshing state... [id=17170099:ecstaticpilot]
github_actions_repository_permissions.devops_collie_actions: Refreshing state... [id=collie]
github_team_membership.cmu-courses_anish: Refreshing state... [id=3643907:ap-1]
github_team_membership.cmu-courses_etashjha: Refreshing state... [id=3643907:etashj]
github_team_membership.cmu-courses_dkou: Refreshing state... [id=3643907:sharkree]
github_team_membership.cmu-courses_gabrielx: Refreshing state... [id=3643907:gabriel-xu]
github_team_membership.occ_xboxbedrock: Refreshing state... [id=19240310:xboxbedrock]
github_team_membership.occ_dcrodrig: Refreshing state... [id=19240310:newhacker1746]
github_team_membership.occ_acchs: Refreshing state... [id=19240310:acchs-at-cmu]
github_team_membership.occ_anish: Refreshing state... [id=19240310:ap-1]
github_team_membership.cmu-eats_arom1a: Refreshing state... [id=16807600:arom1a]
github_team_membership.cmu-insta_etashjha: Refreshing state... [id=19240311:etashj]
github_actions_repository_permissions.cmu-eats_eats_actions: Refreshing state... [id=eats]
github_actions_repository_permissions.devops_kennel_actions: Refreshing state... [id=kennel]
github_repository_deploy_key.devops_kennel_mirror_deploy_key: Refreshing state... [id=kennel:159665328]
github_actions_repository_permissions.cmu-insta_insta_actions: Refreshing state... [id=insta]
github_actions_repository_permissions.cmu-eats_dining_api_actions: Refreshing state... [id=dining-api]
github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Refreshing state... [id=dining-api:161772815]
github_repository_deploy_key.slai_cmugpt_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-surface:159665348]
github_actions_repository_permissions.slai_cmugpt_surface_actions: Refreshing state... [id=cmugpt-surface]
github_actions_repository_permissions.cbp_groupme_mirror_actions: Refreshing state... [id=groupme-mirror]
github_repository_deploy_key.cbp_groupme_mirror_mirror_deploy_key: Refreshing state... [id=groupme-mirror:159665366]
github_actions_repository_permissions.cmu-cal_cal_actions: Refreshing state... [id=cal]
github_repository_deploy_key.cmu-cal_cal_mirror_deploy_key: Refreshing state... [id=cal:159665344]
github_actions_repository_permissions.slai_cmugpt_sms_surface_actions: Refreshing state... [id=cmugpt-sms-surface]
github_repository_deploy_key.slai_cmugpt_sms_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-sms-surface:159665365]
github_actions_repository_permissions.devops_link_shortener_actions: Refreshing state... [id=link-shortener]
github_team_membership.terrier_amzoeee: Refreshing state... [id=12028577:amzoeee]
github_team_membership.terrier_kothurvidur: Refreshing state... [id=12028577:kothurvidur]
github_team_membership.terrier_kritdass: Refreshing state... [id=12028577:kritdass]
github_team_membership.terrier_dkou: Refreshing state... [id=12028577:sharkree]
github_team_membership.terrier_etashjha: Refreshing state... [id=12028577:etashj]
github_team_membership.terrier_anish: Refreshing state... [id=12028577:ap-1]
github_team_membership.devops_xboxbedrock: Refreshing state... [id=14706218:xboxbedrock]
github_team_membership.devops_thesuperrl: Refreshing state... [id=14706218:thesuperrl]
github_team_membership.devops_anish: Refreshing state... [id=14706218:ap-1]
github_actions_repository_permissions.occ_occ_actions: Refreshing state... [id=occ]
github_actions_repository_permissions.devops_infrastructure_actions: Refreshing state... [id=infrastructure]
github_repository_deploy_key.devops_infrastructure_mirror_deploy_key: Refreshing state... [id=infrastructure:159665343]
github_actions_repository_permissions.uia_components_actions: Refreshing state... [id=components]
github_actions_repository_permissions.lost-and-found_lost_and_found_v2_actions: Refreshing state... [id=lost-and-found-v2]
github_repository_deploy_key.lost-and-found_lost_and_found_v2_mirror_deploy_key: Refreshing state... [id=lost-and-found-v2:159665337]
github_actions_repository_permissions.cbp_discord_verify_actions: Refreshing state... [id=discord-verify]
github_repository_deploy_key.cbp_discord_verify_mirror_deploy_key: Refreshing state... [id=discord-verify:159665340]
github_team_membership.cmu-cal_anish: Refreshing state... [id=15669112:ap-1]
github_team_membership.cmu-cal_jalenluorion: Refreshing state... [id=15669112:jalenluorion]
github_actions_repository_permissions.cbp_dalmatian_actions: Refreshing state... [id=dalmatian]
github_team_membership.uia_anish: Refreshing state... [id=17529763:ap-1]
github_team_membership.slai_xboxbedrock: Refreshing state... [id=17571328:xboxbedrock]
github_team_membership.slai_jiyas: Refreshing state... [id=17571328:jiya2301]
github_team_membership.slai_rzhu3: Refreshing state... [id=17571328:richard-31415]
github_team_membership.slai_qianxuege: Refreshing state... [id=17571328:qianxuege]
github_team_membership.slai_krishsax: Refreshing state... [id=17571328:krishsaxena]
github_team_membership.slai_jc_965: Refreshing state... [id=17571328:jc-965]
github_team_membership.slai_anish: Refreshing state... [id=17571328:ap-1]
github_actions_repository_permissions.slai_cmugpt_agent_actions: Refreshing state... [id=cmugpt-agent]
github_repository_deploy_key.slai_cmugpt_agent_mirror_deploy_key: Refreshing state... [id=cmugpt-agent:159665360]
github_actions_repository_permissions.quest_quest_actions: Refreshing state... [id=quest]
github_repository_deploy_key.quest_quest_mirror_deploy_key: Refreshing state... [id=quest:159665323]
github_actions_repository_permissions.uia_scottylabs_org_actions: Refreshing state... [id=scottylabs-org]
github_repository_deploy_key.uia_scottylabs_org_mirror_deploy_key: Refreshing state... [id=scottylabs-org:159665361]
github_actions_repository_permissions.cmu-housing_housing_actions: Refreshing state... [id=housing]
github_repository_deploy_key.cmu-housing_housing_mirror_deploy_key: Refreshing state... [id=housing:159665369]
github_actions_repository_permissions.cmu-courses_courses_actions: Refreshing state... [id=courses]
github_repository_deploy_key.cmu-courses_courses_mirror_deploy_key: Refreshing state... [id=courses:159665327]
github_actions_repository_permissions.terrier_terrier_actions: Refreshing state... [id=terrier]
github_repository_deploy_key.terrier_terrier_mirror_deploy_key: Refreshing state... [id=terrier:159665339]
github_actions_repository_permissions.devops_governance_actions: Refreshing state... [id=governance]
github_repository_deploy_key.devops_governance_mirror_deploy_key: Refreshing state... [id=governance:159665346]
github_team_membership.lost-and-found_anish: Refreshing state... [id=4088608:ap-1]
github_team_membership.cmu-maps_siddharthr: Refreshing state... [id=10872360:overthrowing]
github_team_membership.cmu-maps_anish: Refreshing state... [id=10872360:ap-1]
github_repository_deploy_key.occ_occ_mirror_deploy_key: Refreshing state... [id=occ:161720242]
github_repository_deploy_key.cbp_dalmatian_mirror_deploy_key: Refreshing state... [id=dalmatian:159665367]
github_repository_deploy_key.devops_link_shortener_mirror_deploy_key: Refreshing state... [id=link-shortener:159665345]
github_repository_deploy_key.devops_documentation_mirror_deploy_key: Refreshing state... [id=documentation:159665362]
github_actions_repository_permissions.devops_documentation_actions: Refreshing state... [id=documentation]
github_repository_deploy_key.devops_collie_mirror_deploy_key: Refreshing state... [id=collie:159665332]
github_repository_deploy_key.uia_components_mirror_deploy_key: Refreshing state... [id=components:159665353]
github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Refreshing state... [id=eats:161775995]
github_repository_deploy_key.cmu-insta_insta_mirror_deploy_key: Refreshing state... [id=insta:161720245]
github_repository_deploy_key.tartan-vote_tartan_vote_mirror_deploy_key: Refreshing state... [id=tartan-vote:159665350]
github_actions_repository_permissions.tartan-vote_tartan_vote_actions: Refreshing state... [id=tartan-vote]
forgejo_team_repository.terrier_leads_terrier: Refreshing state...
forgejo_team_repository.housing_leads_housing: Refreshing state...
forgejo_team_repository.scottylabs-org_leads_scottylabs_org: Refreshing state...
forgejo_team_repository.components_leads_components: Refreshing state...
forgejo_team_repository.cal_leads_cal: Refreshing state...
forgejo_team_repository.occ-main_leads_occ: Refreshing state...
forgejo_team_repository.quest_leads_quest: Refreshing state...
forgejo_team_repository.cmugpt-surface_leads_cmugpt_surface: Refreshing state...
forgejo_team_repository.cmugpt-agent_leads_cmugpt_agent: Refreshing state...
forgejo_team_repository.cmugpt-sms-surface_leads_cmugpt_sms_surface: Refreshing state...
forgejo_team_repository.cmugpt-mcp-server_leads_mcp_server: Refreshing state...
forgejo_team_repository.cmu-courses_leads_internet_archive: Refreshing state...
forgejo_team_repository.courses_leads_courses: Refreshing state...
forgejo_team_repository.devops_leads_infrastructure: Refreshing state...
forgejo_team_repository.link-shortener_leads_link_shortener: Refreshing state...
forgejo_team_repository.devops_leads_governance: Refreshing state...
forgejo_team_repository.devops_leads_collie: Refreshing state...
forgejo_team_repository.devops_leads_kennel: Refreshing state...
forgejo_team_repository.documentation_leads_documentation: Refreshing state...
forgejo_team_repository.insta_leads_insta: Refreshing state...
forgejo_team_repository.discord-verify_leads_discord_verify: Refreshing state...
forgejo_team_repository.bus-sign_leads_bus_sign: Refreshing state...
forgejo_team_repository.dalmatian_leads_dalmatian: Refreshing state...
forgejo_team_repository.groupme-mirror_leads_groupme_mirror: Refreshing state...
forgejo_team_repository.tartan-vote_leads_tartan_vote: Refreshing state...
github_repository_deploy_key.cmu-courses_internet_archive_mirror_deploy_key: Refreshing state... [id=internet-archive:159665356]
github_actions_repository_permissions.cmu-courses_internet_archive_actions: Refreshing state... [id=internet-archive]
forgejo_team_repository.maps_leads_maps: Refreshing state...
forgejo_team_repository.lost-and-found-v2_leads_lost_and_found_v2: Refreshing state...
github_team_membership.tartan-vote_maybe_yiyi: Refreshing state... [id=17701833:maybe-yiyi]
github_team_membership.tartan-vote_xboxbedrock: Refreshing state... [id=17701833:xboxbedrock]
github_team_membership.tartan-vote_tentype: Refreshing state... [id=17701833:tentype]
github_team_membership.tartan-vote_riztofu: Refreshing state... [id=17701833:riztofu]
github_team_membership.tartan-vote_anish: Refreshing state... [id=17701833:ap-1]
github_team_membership.cmu-housing_amzoeee: Refreshing state... [id=17937919:amzoeee]
github_team_membership.cmu-housing_xboxbedrock: Refreshing state... [id=17937919:xboxbedrock]
github_team_membership.cmu-housing_nikhilr: Refreshing state... [id=17937919:ecstaticpilot]
github_team_membership.cmu-housing_rhsen: Refreshing state... [id=17937919:rohansen08]
github_team_membership.cmu-housing_jessica_lr_song: Refreshing state... [id=17937919:jessica-lr-song]
github_team_membership.cmu-housing_tentype: Refreshing state... [id=17937919:tentype]
github_team_membership.cmu-housing_stringatlas: Refreshing state... [id=17937919:stringatlas]
github_team_membership.cmu-housing_anish: Refreshing state... [id=17937919:ap-1]
github_team_membership.cmu-housing_emmawong: Refreshing state... [id=17937919:emw868]
github_team_membership.cmu-housing_manrajs: Refreshing state... [id=17937919:manraj10]
github_team_membership.cmu-housing_gostmeaper: Refreshing state... [id=17937919:gostmeaper]
github_team_membership.cmu-housing_kluoxx: Refreshing state... [id=17937919:luokaden888]
github_team_membership.cmu-housing_zoexue: Refreshing state... [id=17937919:o-zox-o]
github_team_membership.quest_xboxbedrock: Refreshing state... [id=17330197:xboxbedrock]
github_team_membership.quest_brightz: Refreshing state... [id=17330197:bright-zheng]
github_team_membership.quest_jw8: Refreshing state... [id=17330197:jeffreyyijiwang]
github_team_membership.quest_kechezon: Refreshing state... [id=17330197:kechezon]
github_team_membership.quest_anish: Refreshing state... [id=17330197:ap-1]
github_actions_repository_permissions.cbp_bus_sign_actions: Refreshing state... [id=bus-sign]
github_repository_deploy_key.cbp_bus_sign_mirror_deploy_key: Refreshing state... [id=bus-sign:159665342]
github_actions_repository_permissions.slai_mcp_server_actions: Refreshing state... [id=mcp-server]
github_repository_deploy_key.slai_mcp_server_mirror_deploy_key: Refreshing state... [id=mcp-server:159665358]
github_repository_deploy_key.cmu-maps_maps_mirror_deploy_key: Refreshing state... [id=maps:159665325]
github_actions_repository_permissions.cmu-maps_maps_actions: Refreshing state... [id=maps]
github_repository_ruleset.devops_kennel_ruleset: Refreshing state... [id=13193453]
github_repository_ruleset.cmu-eats_dining_api_ruleset: Refreshing state... [id=21879626]
github_repository_ruleset.slai_cmugpt_surface_ruleset: Refreshing state... [id=16476662]
github_repository_ruleset.cbp_groupme_mirror_ruleset: Refreshing state... [id=12503296]
github_repository_ruleset.cmu-cal_cal_ruleset: Refreshing state... [id=18919281]
github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Refreshing state... [id=17653622]
github_repository_ruleset.devops_infrastructure_ruleset: Refreshing state... [id=11275118]
github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Refreshing state... [id=18920618]
github_repository_ruleset.cbp_discord_verify_ruleset: Refreshing state... [id=11449154]
github_repository_ruleset.slai_cmugpt_agent_ruleset: Refreshing state... [id=16476661]
github_repository_ruleset.quest_quest_ruleset: Refreshing state... [id=15750052]
github_repository_ruleset.uia_scottylabs_org_ruleset: Refreshing state... [id=18112214]
github_repository_ruleset.cmu-housing_housing_ruleset: Refreshing state... [id=17421604]
github_repository_ruleset.cmu-courses_courses_ruleset: Refreshing state... [id=15855161]
github_repository_ruleset.terrier_terrier_ruleset: Refreshing state... [id=12671003]
github_repository_ruleset.devops_governance_ruleset: Refreshing state... [id=14488792]
github_repository_ruleset.occ_occ_ruleset: Refreshing state... [id=21841634]
github_repository_ruleset.cbp_dalmatian_ruleset: Refreshing state... [id=17238915]
github_repository_ruleset.devops_link_shortener_ruleset: Refreshing state... [id=17911647]
github_repository_ruleset.devops_documentation_ruleset: Refreshing state... [id=17079368]
github_repository_ruleset.devops_collie_ruleset: Refreshing state... [id=18865204]
github_repository_ruleset.uia_components_ruleset: Refreshing state... [id=16366085]
github_repository_ruleset.cmu-eats_eats_ruleset: Refreshing state... [id=21330218]
github_repository_ruleset.cmu-insta_insta_ruleset: Refreshing state... [id=21841636]
github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Refreshing state... [id=16848966]
github_repository_ruleset.cmu-courses_internet_archive_ruleset: Refreshing state... [id=11988074]
github_repository_ruleset.cbp_bus_sign_ruleset: Refreshing state... [id=15236550]
github_repository_ruleset.slai_mcp_server_ruleset: Refreshing state... [id=17082154]
github_repository_ruleset.cmu-maps_maps_ruleset: Refreshing state... [id=18864961]

Note: Objects have changed outside of OpenTofu

OpenTofu detected the following changes made outside of OpenTofu since the
last "tofu apply" which may have affected this plan:

  # restapi_object.cmu-eats_dining_api_push_mirror has been deleted
  - resource "restapi_object" "cmu-eats_dining_api_push_mirror" {
      - api_data        = {
          - "public_key"     = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIIKyjdH3wH8J3aSgeZMLB9/uUGzyYltQsk5Cl1Q8MXqF"
            # (9 unchanged elements hidden)
        } -> null
        id              = "remote_mirror_a4MsQ5rO1_Z"
        # (5 unchanged attributes hidden)
    }

  # restapi_object.cmu-eats_eats_push_mirror has been deleted
  - resource "restapi_object" "cmu-eats_eats_push_mirror" {
      - api_data        = {
          - "public_key"     = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFsm79Zlyi0JKH4y/BdgCCoLyrC4/18yF3owfsqOt53/"
            # (9 unchanged elements hidden)
        } -> null
        id              = "remote_mirror_sjYwMA6NPgh"
        # (5 unchanged attributes hidden)
    }


Unless you have made equivalent changes to your configuration, or ignored the
relevant attributes using ignore_changes, the following plan may include
actions to undo or respond to these changes.

─────────────────────────────────────────────────────────────────────────────

OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
  + create
  ~ update in-place (current -> planned)
-/+ destroy and then create replacement

OpenTofu planned the following actions, but then encountered a problem:

  # forgejo_repository.cmu-eats_dining_api will be updated in-place
  ~ resource "forgejo_repository" "cmu-eats_dining_api" {
      ~ clone_url                   = "https://git.cmu.dev/ScottyLabs/dining-api.git" -> (known after apply)
      ~ created_at                  = "2026-09-18T23:16:27Z" -> (known after apply)
      + description                 = "Dining API for CMU dining locations and availability"
      ~ empty                       = false -> (known after apply)
      + external_tracker            = (known after apply)
      + external_wiki               = (known after apply)
      ~ forks_count                 = 0 -> (known after apply)
      ~ full_name                   = "ScottyLabs/dining-api" -> (known after apply)
      ~ html_url                    = "https://git.cmu.dev/ScottyLabs/dining-api" -> (known after apply)
        id                          = 97
      ~ internal_tracker            = {
          ~ allow_only_contributors_to_track_time = true -> (known after apply)
          ~ enable_issue_dependencies             = true -> (known after apply)
          ~ enable_time_tracker                   = true -> (known after apply)
        } -> (known after apply)
      + mirror_interval             = (known after apply)
      ~ mirror_updated              = "0001-01-01T00:00:00Z" -> (known after apply)
        name                        = "dining-api"
      ~ open_issues_count           = 0 -> (known after apply)
      ~ open_pr_counter             = 5 -> (known after apply)
      ~ permissions                 = {
          ~ admin = true -> (known after apply)
          ~ pull  = true -> (known after apply)
          ~ push  = true -> (known after apply)
        } -> (known after apply)
      ~ release_counter             = 0 -> (known after apply)
      ~ size                        = 38816 -> (known after apply)
      ~ ssh_url                     = "ssh://forgejo@git.cmu.dev/ScottyLabs/dining-api.git" -> (known after apply)
      ~ stars_count                 = 0 -> (known after apply)
      ~ updated_at                  = "2026-09-18T23:22:22Z" -> (known after apply)
      ~ watchers_count              = 3 -> (known after apply)
      ~ website                     = "https://git.cmu.dev/ScottyLabs/dining-api" -> "https://api.cmueats.com"
        # (30 unchanged attributes hidden)
    }

  # forgejo_repository.cmu-eats_eats will be updated in-place
  ~ resource "forgejo_repository" "cmu-eats_eats" {
      ~ clone_url                   = "https://git.cmu.dev/ScottyLabs/eats.git" -> (known after apply)
      ~ created_at                  = "2026-09-18T23:15:39Z" -> (known after apply)
      ~ description                 = "Vite web app for CMU dining location info" -> "CMU dining location info"
      ~ empty                       = false -> (known after apply)
      + external_tracker            = (known after apply)
      + external_wiki               = (known after apply)
      ~ forks_count                 = 0 -> (known after apply)
      ~ full_name                   = "ScottyLabs/eats" -> (known after apply)
      ~ html_url                    = "https://git.cmu.dev/ScottyLabs/eats" -> (known after apply)
        id                          = 96
      ~ internal_tracker            = {
          ~ allow_only_contributors_to_track_time = true -> (known after apply)
          ~ enable_issue_dependencies             = true -> (known after apply)
          ~ enable_time_tracker                   = true -> (known after apply)
        } -> (known after apply)
      + mirror_interval             = (known after apply)
      ~ mirror_updated              = "0001-01-01T00:00:00Z" -> (known after apply)
        name                        = "eats"
      ~ open_issues_count           = 48 -> (known after apply)
      ~ open_pr_counter             = 4 -> (known after apply)
      ~ permissions                 = {
          ~ admin = true -> (known after apply)
          ~ pull  = true -> (known after apply)
          ~ push  = true -> (known after apply)
        } -> (known after apply)
      ~ release_counter             = 0 -> (known after apply)
      ~ size                        = 95496 -> (known after apply)
      ~ ssh_url                     = "ssh://forgejo@git.cmu.dev/ScottyLabs/eats.git" -> (known after apply)
      ~ stars_count                 = 0 -> (known after apply)
      ~ updated_at                  = "2026-09-18T23:35:12Z" -> (known after apply)
      ~ watchers_count              = 3 -> (known after apply)
        # (31 unchanged attributes hidden)
    }

  # forgejo_repository.cmu-study_study will be created
  + resource "forgejo_repository" "cmu-study_study" {
      + allow_manual_merge          = false
      + allow_merge_commits         = true
      + allow_rebase                = true
      + allow_rebase_explicit       = true
      + allow_squash_merge          = true
      + archive_on_destroy          = false
      + archived                    = false
      + auto_init                   = true
      + autodetect_manual_merge     = false
      + avatar_url                  = (known after apply)
      + clone_url                   = (known after apply)
      + created_at                  = (known after apply)
      + default_branch              = "main"
      + default_merge_style         = "merge"
      + description                 = "A platform for creating and joining study groups"
      + empty                       = (known after apply)
      + external_tracker            = (known after apply)
      + external_wiki               = (known after apply)
      + fork                        = (known after apply)
      + forks_count                 = (known after apply)
      + full_name                   = (known after apply)
      + has_actions                 = true
      + has_issues                  = true
      + has_packages                = true
      + has_projects                = true
      + has_pull_requests           = true
      + has_releases                = true
      + has_wiki                    = true
      + html_url                    = (known after apply)
      + id                          = (known after apply)
      + ignore_whitespace_conflicts = false
      + internal                    = (known after apply)
      + internal_tracker            = (known after apply)
      + labels                      = false
      + lfs                         = false
      + milestones                  = false
      + mirror                      = false
      + mirror_interval             = (known after apply)
      + mirror_updated              = (known after apply)
      + name                        = "study"
      + open_issues_count           = (known after apply)
      + open_pr_counter             = (known after apply)
      + owner                       = "ScottyLabs"
      + parent_id                   = (known after apply)
      + permissions                 = (known after apply)
      + private                     = false
      + release_counter             = (known after apply)
      + size                        = (known after apply)
      + ssh_url                     = (known after apply)
      + stars_count                 = (known after apply)
      + template                    = false
      + trust_model                 = "default"
      + updated_at                  = (known after apply)
      + watchers_count              = (known after apply)
      + website                     = "https://cmustudy.com"
    }

  # forgejo_team.cmu-study will be created
  + resource "forgejo_team" "cmu-study" {
      + can_create_org_repo       = (known after apply)
      + description               = "CMU Study"
      + id                        = (known after apply)
      + includes_all_repositories = (known after apply)
      + name                      = "cmu-study"
      + organization              = (known after apply)
      + organization_id           = 3
      + permission                = "write"
      + units_map                 = {
          + "repo.code"   = "write"
          + "repo.issues" = "write"
          + "repo.pulls"  = "write"
        }
    }

  # forgejo_team.study_leads will be created
  + resource "forgejo_team" "study_leads" {
      + can_create_org_repo       = false
      + description               = "CMU Study leads"
      + id                        = (known after apply)
      + includes_all_repositories = false
      + name                      = "study-leads"
      + organization              = (known after apply)
      + organization_id           = 3
      + permission                = "admin"
      + units_map                 = {
          + "repo.actions"    = "admin"
          + "repo.code"       = "admin"
          + "repo.ext_issues" = "admin"
          + "repo.ext_wiki"   = "admin"
          + "repo.issues"     = "admin"
          + "repo.packages"   = "admin"
          + "repo.projects"   = "admin"
          + "repo.pulls"      = "admin"
          + "repo.releases"   = "admin"
          + "repo.wiki"       = "admin"
        }
    }

  # forgejo_team_member.cmu-study_annadavi will be created
  + resource "forgejo_team_member" "cmu-study_annadavi" {
      + team_id = (known after apply)
      + user    = "annadavi"
    }

  # forgejo_team_member.study_leads_annadavi will be created
  + resource "forgejo_team_member" "study_leads_annadavi" {
      + team_id = (known after apply)
      + user    = "annadavi"
    }

  # forgejo_team_repository.cmu-study_study will be created
  + resource "forgejo_team_repository" "cmu-study_study" {
      + owner      = "ScottyLabs"
      + repository = "study"
      + team_id    = (known after apply)
    }

  # forgejo_team_repository.study_leads_study will be created
  + resource "forgejo_team_repository" "study_leads_study" {
      + owner      = "ScottyLabs"
      + repository = "study"
      + team_id    = (known after apply)
    }

  # github_actions_repository_permissions.cmu-study_study_actions will be created
  + resource "github_actions_repository_permissions" "cmu-study_study_actions" {
      + enabled              = false
      + id                   = (known after apply)
      + repository           = "study"
      + sha_pinning_required = (known after apply)
    }

  # github_repository.cmu-study_study will be created
  + resource "github_repository" "cmu-study_study" {
      + allow_auto_merge                        = false
      + allow_forking                           = (known after apply)
      + allow_merge_commit                      = true
      + allow_rebase_merge                      = true
      + allow_squash_merge                      = true
      + archived                                = false
      + default_branch                          = (known after apply)
      + delete_branch_on_merge                  = false
      + description                             = "READ-ONLY MIRROR"
      + etag                                    = (known after apply)
      + fork                                    = (known after apply)
      + full_name                               = (known after apply)
      + git_clone_url                           = (known after apply)
      + has_discussions                         = false
      + has_issues                              = false
      + has_projects                            = false
      + has_wiki                                = false
      + homepage_url                            = "https://git.cmu.dev/ScottyLabs/study"
      + html_url                                = (known after apply)
      + http_clone_url                          = (known after apply)
      + id                                      = (known after apply)
      + ignore_vulnerability_alerts_during_read = false
      + merge_commit_message                    = "PR_TITLE"
      + merge_commit_title                      = "MERGE_MESSAGE"
      + name                                    = "study"
      + node_id                                 = (known after apply)
      + primary_language                        = (known after apply)
      + private                                 = (known after apply)
      + repo_id                                 = (known after apply)
      + source_owner                            = (known after apply)
      + source_repo                             = (known after apply)
      + squash_merge_commit_message             = "COMMIT_MESSAGES"
      + squash_merge_commit_title               = "COMMIT_OR_PR_TITLE"
      + ssh_clone_url                           = (known after apply)
      + svn_url                                 = (known after apply)
      + topics                                  = (known after apply)
      + visibility                              = "public"
      + vulnerability_alerts                    = (known after apply)
      + web_commit_signoff_required             = (known after apply)

      + security_and_analysis (known after apply)
    }

  # github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key must be replaced
-/+ resource "github_repository_deploy_key" "cmu-eats_dining_api_mirror_deploy_key" {
      ~ etag       = "W/\"bf394a0fcc9e1231b4ee544598e6680359d71d177764d9f0c8cd22f5c7ec3ef4\"" -> (known after apply)
      ~ id         = "dining-api:161772815" -> (known after apply)
      ~ key        = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIIKyjdH3wH8J3aSgeZMLB9/uUGzyYltQsk5Cl1Q8MXqF" # forces replacement -> (known after apply) # forces replacement
        # (3 unchanged attributes hidden)
    }

  # github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key must be replaced
-/+ resource "github_repository_deploy_key" "cmu-eats_eats_mirror_deploy_key" {
      ~ etag       = "W/\"5cf445e3b52f1b2e59fe1cb32c5c2e0859e2598566a7aa75c3ebdbf4b01f6164\"" -> (known after apply)
      ~ id         = "eats:161775995" -> (known after apply)
      ~ key        = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFsm79Zlyi0JKH4y/BdgCCoLyrC4/18yF3owfsqOt53/" # forces replacement -> (known after apply) # forces replacement
        # (3 unchanged attributes hidden)
    }

  # github_repository_deploy_key.cmu-study_study_mirror_deploy_key will be created
  + resource "github_repository_deploy_key" "cmu-study_study_mirror_deploy_key" {
      + etag       = (known after apply)
      + id         = (known after apply)
      + key        = (known after apply)
      + read_only  = false
      + repository = "study"
      + title      = "Forgejo mirror"
    }

  # github_repository_ruleset.cmu-eats_dining_api_ruleset will be updated in-place
  ~ resource "github_repository_ruleset" "cmu-eats_dining_api_ruleset" {
        id          = "21879626"
        name        = "Default"
        # (6 unchanged attributes hidden)

      ~ bypass_actors {
          ~ actor_id    = 161772815 -> (known after apply)
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place
  ~ resource "github_repository_ruleset" "cmu-eats_eats_ruleset" {
        id          = "21330218"
        name        = "Default"
        # (6 unchanged attributes hidden)

      ~ bypass_actors {
          ~ actor_id    = 161775995 -> (known after apply)
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.cmu-study_study_ruleset will be created
  + resource "github_repository_ruleset" "cmu-study_study_ruleset" {
      + enforcement = "active"
      + etag        = (known after apply)
      + id          = (known after apply)
      + name        = "Default"
      + node_id     = (known after apply)
      + repository  = "study"
      + ruleset_id  = (known after apply)
      + target      = "branch"

      + bypass_actors {
          + actor_id    = (known after apply)
          + actor_type  = "DeployKey"
          + bypass_mode = "always"
        }

      + conditions {
          + ref_name {
              + exclude = []
              + include = [
                  + "~ALL",
                ]
            }
        }

      + rules {
          + creation                      = true
          + deletion                      = true
          + non_fast_forward              = true
          + update                        = true
          + update_allows_fetch_and_merge = false

          + pull_request {
              + allowed_merge_methods             = (known after apply)
              + dismiss_stale_reviews_on_push     = false
              + require_code_owner_review         = false
              + require_last_push_approval        = false
              + required_approving_review_count   = 1
              + required_review_thread_resolution = false
            }
        }
    }

  # github_team.cmu-study will be created
  + resource "github_team" "cmu-study" {
      + create_default_maintainer = false
      + etag                      = (known after apply)
      + id                        = (known after apply)
      + members_count             = (known after apply)
      + name                      = "CMU Study"
      + node_id                   = (known after apply)
      + notification_setting      = "notifications_enabled"
      + parent_team_read_id       = (known after apply)
      + parent_team_read_slug     = (known after apply)
      + privacy                   = "closed"
      + slug                      = (known after apply)
    }

  # github_team_membership.cmu-study_annadavi will be created
  + resource "github_team_membership" "cmu-study_annadavi" {
      + etag     = (known after apply)
      + id       = (known after apply)
      + role     = "maintainer"
      + team_id  = (known after apply)
      + username = "annadavid1"
    }

  # restapi_object.cmu-eats_dining_api_push_mirror will be created
  + resource "restapi_object" "cmu-eats_dining_api_push_mirror" {
      + api_data        = (known after apply)
      + api_response    = (known after apply)
      + create_response = (known after apply)
      + data            = jsonencode(
            {
              + interval       = "8h0m0s"
              + remote_address = "git@github.com:ScottyLabs/dining-api.git"
              + sync_on_commit = true
              + use_ssh        = true
            }
        )
      + id              = (known after apply)
      + id_attribute    = "remote_name"
      + path            = "/api/v1/repos/ScottyLabs/dining-api/push_mirrors"
    }

  # restapi_object.cmu-eats_eats_kennel_webhook will be created
  + resource "restapi_object" "cmu-eats_eats_kennel_webhook" {
      + api_data        = (known after apply)
      + api_response    = (known after apply)
      + create_response = (known after apply)
      + data            = (sensitive value)
      + id              = (known after apply)
      + id_attribute    = "id"
      + path            = "/api/v1/repos/ScottyLabs/eats/hooks"
      + update_method   = "PATCH"
    }

  # restapi_object.cmu-eats_eats_push_mirror will be created
  + resource "restapi_object" "cmu-eats_eats_push_mirror" {
      + api_data        = (known after apply)
      + api_response    = (known after apply)
      + create_response = (known after apply)
      + data            = jsonencode(
            {
              + interval       = "8h0m0s"
              + remote_address = "git@github.com:ScottyLabs/eats.git"
              + sync_on_commit = true
              + use_ssh        = true
            }
        )
      + id              = (known after apply)
      + id_attribute    = "remote_name"
      + path            = "/api/v1/repos/ScottyLabs/eats/push_mirrors"
    }

  # restapi_object.cmu-study_study_kennel_webhook will be created
  + resource "restapi_object" "cmu-study_study_kennel_webhook" {
      + api_data        = (known after apply)
      + api_response    = (known after apply)
      + create_response = (known after apply)
      + data            = (sensitive value)
      + id              = (known after apply)
      + id_attribute    = "id"
      + path            = "/api/v1/repos/ScottyLabs/study/hooks"
      + update_method   = "PATCH"
    }

  # restapi_object.cmu-study_study_push_mirror will be created
  + resource "restapi_object" "cmu-study_study_push_mirror" {
      + api_data        = (known after apply)
      + api_response    = (known after apply)
      + create_response = (known after apply)
      + data            = jsonencode(
            {
              + interval       = "8h0m0s"
              + remote_address = "git@github.com:ScottyLabs/study.git"
              + sync_on_commit = true
              + use_ssh        = true
            }
        )
      + id              = (known after apply)
      + id_attribute    = "remote_name"
      + path            = "/api/v1/repos/ScottyLabs/study/push_mirrors"
    }

Plan: 20 to add, 4 to change, 2 to destroy.
╷
│ Error: Unable to read team repository
│ 
│   with forgejo_team_repository.cmu-eats_dining_api,
│   on forgejo_team_repos.tf.json line 102, in resource.forgejo_team_repository.cmu-eats_dining_api:
│  102:       },
│ 
│ Repository 'ScottyLabs/dining-api' not found in team with ID 28
╵
╷
│ Error: Unable to read team repository
│ 
│   with forgejo_team_repository.cmu-eats_eats,
│   on forgejo_team_repos.tf.json line 111, in resource.forgejo_team_repository.cmu-eats_eats:
│  111:       },
│ 
│ Repository 'ScottyLabs/eats' not found in team with ID 28
╵


Ran Plan for dir: `tofu/git` workspace: `default` **Plan Error** <details><summary>Show Output</summary> ``` running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu plan -input=false -refresh -out /var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git/git-default.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git': exit status 1 Finished `release` profile [optimized] target(s) in 0.05s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ random_password.kennel_webhook_secret: Refreshing state... [id=none] forgejo_repository.cmu-courses_courses: Refreshing state... [name=courses] forgejo_repository.devops_governance: Refreshing state... [name=governance] forgejo_repository.cmu-eats_dining_api: Refreshing state... [name=dining-api] forgejo_repository.cmu-maps_maps: Refreshing state... [name=maps] forgejo_repository.cmu-courses_internet_archive: Refreshing state... [name=internet-archive] forgejo_repository.quest_quest: Refreshing state... [name=quest] forgejo_repository.uia_scottylabs_org: Refreshing state... [name=scottylabs-org] forgejo_repository.devops_infrastructure: Refreshing state... [name=infrastructure] forgejo_repository.cmu-housing_housing: Refreshing state... [name=housing] forgejo_repository.slai_cmugpt_surface: Refreshing state... [name=cmugpt-surface] forgejo_repository.terrier_terrier: Refreshing state... [name=terrier] forgejo_repository.cbp_groupme_mirror: Refreshing state... [name=groupme-mirror] data.forgejo_organization.this: Reading... forgejo_repository.devops_kennel: Refreshing state... [name=kennel] forgejo_repository.cmu-cal_cal: Refreshing state... [name=cal] forgejo_repository.tartan-vote_tartan_vote: Refreshing state... [name=tartan-vote] data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs] forgejo_repository.cbp_discord_verify: Refreshing state... [name=discord-verify] forgejo_repository.lost-and-found_lost_and_found_v2: Refreshing state... [name=lost-and-found-v2] forgejo_repository.slai_cmugpt_sms_surface: Refreshing state... [name=cmugpt-sms-surface] forgejo_repository.slai_mcp_server: Refreshing state... [name=mcp-server] forgejo_repository.slai_cmugpt_agent: Refreshing state... [name=cmugpt-agent] forgejo_repository.occ_occ: Refreshing state... [name=occ] forgejo_repository.cbp_dalmatian: Refreshing state... [name=dalmatian] forgejo_repository.devops_documentation: Refreshing state... [name=documentation] forgejo_repository.devops_collie: Refreshing state... [name=collie] forgejo_repository.devops_link_shortener: Refreshing state... [name=link-shortener] forgejo_repository.uia_components: Refreshing state... [name=components] forgejo_repository.cbp_bus_sign: Refreshing state... [name=bus-sign] forgejo_repository.cmu-eats_eats: Refreshing state... [name=eats] forgejo_repository.cmu-insta_insta: Refreshing state... [name=insta] data.external.identity_riztofu: Reading... data.external.identity_kechezon: Reading... data.external.identity_dcrodrig: Reading... data.external.identity_acchs: Reading... data.external.identity_gostmeaper: Reading... data.external.identity_liamwalk: Reading... data.external.identity_etashjha: Reading... data.external.identity_kluoxx: Reading... data.external.identity_xboxbedrock: Reading... data.external.identity_gabrielx: Reading... data.external.identity_riztofu: Read complete after 0s [id=-] data.external.identity_jw8: Reading... data.external.identity_acchs: Read complete after 0s [id=-] data.external.identity_tentype: Reading... data.external.identity_kechezon: Read complete after 0s [id=-] data.external.identity_brightz: Reading... data.external.identity_dcrodrig: Read complete after 0s [id=-] data.external.identity_jalenluorion: Reading... data.external.identity_etashjha: Read complete after 0s [id=-] data.external.identity_jessica_lr_song: Reading... data.external.identity_liamwalk: Read complete after 0s [id=-] data.external.identity_gabrielx: Read complete after 0s [id=-] data.external.identity_amzoeee: Reading... data.external.identity_zoexue: Reading... data.external.identity_xboxbedrock: Read complete after 0s [id=-] data.external.identity_jc_965: Reading... data.external.identity_kluoxx: Read complete after 0s [id=-] data.external.identity_anish: Reading... data.external.identity_gostmeaper: Read complete after 0s [id=-] data.external.identity_arom1a: Reading... data.external.identity_jw8: Read complete after 0s [id=-] data.external.identity_jiyas: Reading... data.external.identity_tentype: Read complete after 0s [id=-] data.external.identity_krishsax: Reading... data.external.identity_jessica_lr_song: Read complete after 0s [id=-] data.external.identity_jalenluorion: Read complete after 0s [id=-] data.external.identity_annadavi: Reading... data.external.identity_qianxuege: Reading... data.external.identity_brightz: Read complete after 0s [id=-] data.external.identity_maybe_yiyi: Reading... data.external.identity_amzoeee: Read complete after 0s [id=-] data.external.identity_kritdass: Reading... data.external.identity_zoexue: Read complete after 0s [id=-] data.external.identity_dkou: Reading... data.external.identity_jc_965: Read complete after 0s [id=-] data.external.identity_siddharthr: Reading... data.external.identity_anish: Read complete after 0s [id=-] data.external.identity_manrajs: Reading... data.external.identity_arom1a: Read complete after 0s [id=-] data.external.identity_kothurvidur: Reading... data.external.identity_jiyas: Read complete after 1s [id=-] data.external.identity_emmawong: Reading... data.external.identity_maybe_yiyi: Read complete after 1s [id=-] data.external.identity_thesuperrl: Reading... data.external.identity_krishsax: Read complete after 1s [id=-] data.external.identity_annadavi: Read complete after 1s [id=-] data.external.identity_stringatlas: Reading... data.external.identity_rzhu3: Reading... data.external.identity_qianxuege: Read complete after 1s [id=-] data.external.identity_nikhilr: Reading... data.external.identity_dkou: Read complete after 1s [id=-] data.external.identity_rhsen: Reading... data.external.identity_kritdass: Read complete after 1s [id=-] restapi_object.cmu-eats_dining_api_push_mirror: Refreshing state... [id=remote_mirror_a4MsQ5rO1_Z] data.external.identity_siddharthr: Read complete after 1s [id=-] restapi_object.devops_governance_kennel_webhook: Refreshing state... [id=22] data.external.identity_manrajs: Read complete after 1s [id=-] restapi_object.devops_governance_push_mirror: Refreshing state... [id=remote_mirror_lt9JzkX1jMq] data.external.identity_kothurvidur: Read complete after 1s [id=-] restapi_object.devops_infrastructure_push_mirror: Refreshing state... [id=remote_mirror_oYtFpKVzTaZ] restapi_object.devops_infrastructure_kennel_webhook: Refreshing state... [id=26] restapi_object.cmu-courses_courses_push_mirror: Refreshing state... [id=remote_mirror_WMR9yQwKFDF] restapi_object.cmu-courses_courses_kennel_webhook: Refreshing state... [id=11] restapi_object.cmu-maps_maps_push_mirror: Refreshing state... [id=remote_mirror_e4_FUnvoslJ] restapi_object.cmu-maps_maps_kennel_webhook: Refreshing state... [id=9] restapi_object.uia_scottylabs_org_kennel_webhook: Refreshing state... [id=29] restapi_object.cmu-courses_internet_archive_push_mirror: Refreshing state... [id=remote_mirror_58X__Yfs0bR] restapi_object.uia_scottylabs_org_push_mirror: Refreshing state... [id=remote_mirror_h6yN6QdS1lD] restapi_object.quest_quest_kennel_webhook: Refreshing state... [id=10] restapi_object.quest_quest_push_mirror: Refreshing state... [id=remote_mirror_U9d9BPAon29] data.external.identity_emmawong: Read complete after 0s [id=-] forgejo_team.terrier: Refreshing state... [name=terrier] forgejo_team.cmugpt-sms-surface_leads: Refreshing state... [name=cmugpt-sms-surface-leads] forgejo_team.cmu-eats: Refreshing state... [name=cmu-eats] forgejo_team.bus-sign_leads: Refreshing state... [name=bus-sign-leads] forgejo_team.cmu-housing: Refreshing state... [name=cmu-housing] data.external.identity_stringatlas: Read complete after 0s [id=-] forgejo_team.cmu-courses_leads: Refreshing state... [name=cmu-courses-leads] forgejo_team.documentation_leads: Refreshing state... [name=documentation-leads] data.external.identity_rhsen: Read complete after 0s [id=-] data.external.identity_nikhilr: Read complete after 0s [id=-] data.external.identity_rzhu3: Read complete after 0s [id=-] forgejo_team.devops_leads: Refreshing state... [name=devops-leads] forgejo_team.uia: Refreshing state... [name=uia] forgejo_team.scottylabs-org_leads: Refreshing state... [name=scottylabs-org-leads] forgejo_team.cmu-cal: Refreshing state... [name=cmu-cal] forgejo_team.eats_leads: Refreshing state... [name=eats-leads] forgejo_team.quest_leads: Refreshing state... [name=quest-leads] forgejo_team.cmu-courses: Refreshing state... [name=cmu-courses] forgejo_team.occ: Refreshing state... [name=occ] data.external.identity_thesuperrl: Read complete after 0s [id=-] forgejo_team.occ-main_leads: Refreshing state... [name=occ-main-leads] forgejo_team.quest: Refreshing state... [name=quest] forgejo_team.lost-and-found-v2_leads: Refreshing state... [name=lost-and-found-v2-leads] forgejo_team.slai: Refreshing state... [name=slai] forgejo_team.tartan-vote_leads: Refreshing state... [name=tartan-vote-leads] forgejo_team.courses_leads: Refreshing state... [name=courses-leads] forgejo_team.devops: Refreshing state... [name=devops] forgejo_team.cmugpt-agent_leads: Refreshing state... [name=cmugpt-agent-leads] forgejo_team.cmu-insta: Refreshing state... [name=cmu-insta] forgejo_team.maps_leads: Refreshing state... [name=maps-leads] forgejo_team.dalmatian_leads: Refreshing state... [name=dalmatian-leads] forgejo_team.insta_leads: Refreshing state... [name=insta-leads] forgejo_team.tartan-vote: Refreshing state... [name=tartan-vote] forgejo_team.terrier_leads: Refreshing state... [name=terrier-leads] forgejo_team.cbp: Refreshing state... [name=cbp] forgejo_team.discord-verify_leads: Refreshing state... [name=discord-verify-leads] forgejo_team.cmu-maps: Refreshing state... [name=cmu-maps] forgejo_team.components_leads: Refreshing state... [name=components-leads] forgejo_team.link-shortener_leads: Refreshing state... [name=link-shortener-leads] forgejo_team.cal_leads: Refreshing state... [name=cal-leads] forgejo_team.lost-and-found: Refreshing state... [name=lost-and-found] forgejo_team.cmugpt-surface_leads: Refreshing state... [name=cmugpt-surface-leads] forgejo_team.housing_leads: Refreshing state... [name=housing-leads] forgejo_team.cmugpt-mcp-server_leads: Refreshing state... [name=cmugpt-mcp-server-leads] forgejo_team.groupme-mirror_leads: Refreshing state... [name=groupme-mirror-leads] restapi_object.terrier_terrier_push_mirror: Refreshing state... [id=remote_mirror_YopDgZksZv5] restapi_object.terrier_terrier_kennel_webhook: Refreshing state... [id=12] restapi_object.slai_cmugpt_surface_kennel_webhook: Refreshing state... [id=23] restapi_object.slai_cmugpt_surface_push_mirror: Refreshing state... [id=remote_mirror_cPTQN6S5SpB] restapi_object.cmu-housing_housing_push_mirror: Refreshing state... [id=remote_mirror_uoV2HG4aiOT] restapi_object.cmu-housing_housing_kennel_webhook: Refreshing state... [id=16] restapi_object.cmu-housing_housing_docs_webhook: Refreshing state... [id=15] restapi_object.tartan-vote_tartan_vote_push_mirror: Refreshing state... [id=remote_mirror_EUFq4sFSnH5] restapi_object.tartan-vote_tartan_vote_kennel_webhook: Refreshing state... [id=24] restapi_object.tartan-vote_tartan_vote_docs_webhook: Refreshing state... [id=25] restapi_object.devops_kennel_push_mirror: Refreshing state... [id=remote_mirror_cT9psJkhTQ4] restapi_object.devops_kennel_kennel_webhook: Refreshing state... [id=14] restapi_object.cbp_groupme_mirror_push_mirror: Refreshing state... [id=remote_mirror_U61HJHB9VGE] restapi_object.lost-and-found_lost_and_found_v2_push_mirror: Refreshing state... [id=remote_mirror_n_UZBFU-CY4] restapi_object.cmu-cal_cal_kennel_webhook: Refreshing state... [id=20] restapi_object.cmu-cal_cal_push_mirror: Refreshing state... [id=remote_mirror_CO243uIONoq] restapi_object.cbp_discord_verify_kennel_webhook: Refreshing state... [id=18] restapi_object.cbp_discord_verify_push_mirror: Refreshing state... [id=remote_mirror_3MV7Up6cY1X] restapi_object.slai_cmugpt_sms_surface_push_mirror: Refreshing state... [id=remote_mirror__fkBkASN2jA] restapi_object.slai_cmugpt_sms_surface_kennel_webhook: Refreshing state... [id=19] restapi_object.slai_mcp_server_push_mirror: Refreshing state... [id=remote_mirror_Tqh2y0R_vDp] restapi_object.slai_mcp_server_kennel_webhook: Refreshing state... [id=13] restapi_object.slai_cmugpt_agent_push_mirror: Refreshing state... [id=remote_mirror_kNY6rVcxdyr] restapi_object.slai_cmugpt_agent_kennel_webhook: Refreshing state... [id=28] github_team.cbp: Refreshing state... [id=17170099] github_repository.cmu-eats_eats: Refreshing state... [id=eats] github_repository.devops_kennel: Refreshing state... [id=kennel] github_repository.devops_collie: Refreshing state... [id=collie] github_repository.cmu-insta_insta: Refreshing state... [id=insta] github_repository.cmu-eats_dining_api: Refreshing state... [id=dining-api] github_team.cmu-courses: Refreshing state... [id=3643907] github_team.occ: Refreshing state... [id=19240310] github_team.cmu-eats: Refreshing state... [id=16807600] github_team.cmu-insta: Refreshing state... [id=19240311] github_repository.cmu-cal_cal: Refreshing state... [id=cal] github_repository.devops_link_shortener: Refreshing state... [id=link-shortener] github_repository.occ_occ: Refreshing state... [id=occ] github_repository.devops_infrastructure: Refreshing state... [id=infrastructure] github_repository.slai_cmugpt_surface: Refreshing state... [id=cmugpt-surface] github_repository.cbp_groupme_mirror: Refreshing state... [id=groupme-mirror] github_repository.slai_cmugpt_sms_surface: Refreshing state... [id=cmugpt-sms-surface] github_repository.cbp_dalmatian: Refreshing state... [id=dalmatian] github_team.terrier: Refreshing state... [id=12028577] github_team.devops: Refreshing state... [id=14706218] github_repository.uia_components: Refreshing state... [id=components] github_repository.lost-and-found_lost_and_found_v2: Refreshing state... [id=lost-and-found-v2] github_repository.cbp_discord_verify: Refreshing state... [id=discord-verify] github_team.cmu-cal: Refreshing state... [id=15669112] github_repository.quest_quest: Refreshing state... [id=quest] github_repository.uia_scottylabs_org: Refreshing state... [id=scottylabs-org] github_repository.cmu-housing_housing: Refreshing state... [id=housing] github_team.uia: Refreshing state... [id=17529763] github_repository.terrier_terrier: Refreshing state... [id=terrier] github_repository.devops_governance: Refreshing state... [id=governance] github_team.slai: Refreshing state... [id=17571328] github_repository.slai_cmugpt_agent: Refreshing state... [id=cmugpt-agent] github_repository.devops_documentation: Refreshing state... [id=documentation] github_repository.tartan-vote_tartan_vote: Refreshing state... [id=tartan-vote] github_repository.cmu-courses_courses: Refreshing state... [id=courses] github_team.lost-and-found: Refreshing state... [id=4088608] github_team.cmu-maps: Refreshing state... [id=10872360] github_repository.cbp_bus_sign: Refreshing state... [id=bus-sign] github_repository.slai_mcp_server: Refreshing state... [id=mcp-server] github_repository.cmu-courses_internet_archive: Refreshing state... [id=internet-archive] github_repository.cmu-maps_maps: Refreshing state... [id=maps] github_team.tartan-vote: Refreshing state... [id=17701833] github_team.cmu-housing: Refreshing state... [id=17937919] github_team.quest: Refreshing state... [id=17330197] restapi_object.devops_documentation_push_mirror: Refreshing state... [id=remote_mirror_izmsqaNB6Wx] restapi_object.occ_occ_push_mirror: Refreshing state... [id=remote_mirror_moBLq1YG1qz] restapi_object.cbp_dalmatian_kennel_webhook: Refreshing state... [id=6] restapi_object.cbp_dalmatian_push_mirror: Refreshing state... [id=remote_mirror_qzlCpLK153F] restapi_object.cbp_dalmatian_docs_webhook: Refreshing state... [id=7] restapi_object.devops_link_shortener_push_mirror: Refreshing state... [id=remote_mirror_cp8daMpt82v] restapi_object.devops_link_shortener_kennel_webhook: Refreshing state... [id=8] restapi_object.devops_collie_push_mirror: Refreshing state... [id=remote_mirror_ljvVejNFXaS] restapi_object.devops_collie_kennel_webhook: Refreshing state... [id=17] restapi_object.uia_components_push_mirror: Refreshing state... [id=remote_mirror_Kwwa8TDnTgc] restapi_object.uia_components_kennel_webhook: Refreshing state... [id=27] restapi_object.cmu-eats_eats_kennel_webhook: Refreshing state... [id=31] restapi_object.cmu-eats_eats_push_mirror: Refreshing state... [id=remote_mirror_sjYwMA6NPgh] restapi_object.cmu-insta_insta_push_mirror: Refreshing state... [id=remote_mirror_891nu6XQ1us] restapi_object.cbp_bus_sign_kennel_webhook: Refreshing state... [id=21] restapi_object.cbp_bus_sign_push_mirror: Refreshing state... [id=remote_mirror_MkKxQ2vA9pc] forgejo_team_member.cmugpt-sms-surface_leads_krishsax: Refreshing state... forgejo_team_member.cmugpt-sms-surface_leads_anish: Refreshing state... forgejo_team_member.terrier_amzoeee: Refreshing state... forgejo_team_member.terrier_dkou: Refreshing state... forgejo_team_member.terrier_anish: Refreshing state... forgejo_team_member.terrier_etashjha: Refreshing state... forgejo_team_member.terrier_kritdass: Refreshing state... forgejo_team_member.terrier_kothurvidur: Refreshing state... forgejo_team_repository.terrier_terrier: Refreshing state... forgejo_team_member.cmu-housing_rhsen: Refreshing state... forgejo_team_member.cmu-housing_kluoxx: Refreshing state... forgejo_team_member.cmu-housing_nikhilr: Refreshing state... forgejo_team_member.cmu-housing_stringatlas: Refreshing state... forgejo_team_member.cmu-housing_manrajs: Refreshing state... forgejo_team_member.cmu-housing_amzoeee: Refreshing state... forgejo_team_member.cmu-housing_jessica_lr_song: Refreshing state... forgejo_team_member.cmu-housing_anish: Refreshing state... forgejo_team_member.cmu-housing_tentype: Refreshing state... forgejo_team_member.bus-sign_leads_maybe_yiyi: Refreshing state... forgejo_team_member.bus-sign_leads_anish: Refreshing state... forgejo_team_member.bus-sign_leads_tentype: Refreshing state... forgejo_team_member.cmu-housing_emmawong: Refreshing state... forgejo_team_member.cmu-housing_gostmeaper: Refreshing state... forgejo_team_member.cmu-housing_zoexue: Refreshing state... forgejo_team_repository.cmu-housing_housing: Refreshing state... forgejo_team_repository.cmu-eats_dining_api: Refreshing state... forgejo_team_member.cmu-housing_xboxbedrock: Refreshing state... forgejo_team_repository.cmu-eats_eats: Refreshing state... forgejo_team_member.cmu-eats_arom1a: Refreshing state... forgejo_team_member.cmu-courses_leads_etashjha: Refreshing state... forgejo_team_member.devops_leads_anish: Refreshing state... forgejo_team_member.cmu-courses_leads_anish: Refreshing state... forgejo_team_member.documentation_leads_anish: Refreshing state... forgejo_team_repository.uia_scottylabs_org: Refreshing state... forgejo_team_member.uia_anish: Refreshing state... forgejo_team_member.documentation_leads_thesuperrl: Refreshing state... forgejo_team_member.devops_leads_thesuperrl: Refreshing state... forgejo_team_repository.uia_components: Refreshing state... forgejo_team_member.scottylabs-org_leads_anish: Refreshing state... forgejo_team_member.cmu-cal_anish: Refreshing state... forgejo_team_member.cmu-cal_jalenluorion: Refreshing state... forgejo_team_repository.cmu-cal_cal: Refreshing state... forgejo_team_member.occ_dcrodrig: Refreshing state... forgejo_team_member.occ_anish: Refreshing state... forgejo_team_member.occ_xboxbedrock: Refreshing state... forgejo_team_repository.occ_occ: Refreshing state... forgejo_team_member.occ_acchs: Refreshing state... forgejo_team_member.lost-and-found-v2_leads_anish: Refreshing state... forgejo_team_member.quest_xboxbedrock: Refreshing state... forgejo_team_member.quest_kechezon: Refreshing state... forgejo_team_member.quest_jw8: Refreshing state... forgejo_team_member.quest_brightz: Refreshing state... forgejo_team_repository.quest_quest: Refreshing state... forgejo_team_member.quest_anish: Refreshing state... forgejo_team_member.slai_jc_965: Refreshing state... forgejo_team_repository.slai_cmugpt_surface: Refreshing state... forgejo_team_member.slai_anish: Refreshing state... forgejo_team_member.slai_qianxuege: Refreshing state... forgejo_team_repository.slai_cmugpt_sms_surface: Refreshing state... forgejo_team_repository.slai_cmugpt_agent: Refreshing state... forgejo_team_member.slai_krishsax: Refreshing state... forgejo_team_repository.slai_mcp_server: Refreshing state... forgejo_team_member.slai_rzhu3: Refreshing state... forgejo_team_member.slai_xboxbedrock: Refreshing state... forgejo_team_member.slai_jiyas: Refreshing state... forgejo_team_member.cmu-courses_gabrielx: Refreshing state... forgejo_team_member.cmu-courses_dkou: Refreshing state... forgejo_team_member.cmu-courses_etashjha: Refreshing state... forgejo_team_repository.cmu-courses_internet_archive: Refreshing state... forgejo_team_member.cmu-courses_anish: Refreshing state... forgejo_team_repository.cmu-courses_courses: Refreshing state... forgejo_team_member.eats_leads_arom1a: Refreshing state... forgejo_team_member.occ-main_leads_anish: Refreshing state... forgejo_team_member.occ-main_leads_dcrodrig: Refreshing state... forgejo_team_member.courses_leads_anish: Refreshing state... forgejo_team_member.courses_leads_etashjha: Refreshing state... forgejo_team_repository.devops_infrastructure: Refreshing state... forgejo_team_repository.devops_governance: Refreshing state... forgejo_team_repository.devops_link_shortener: Refreshing state... forgejo_team_member.devops_anish: Refreshing state... forgejo_team_repository.devops_collie: Refreshing state... forgejo_team_repository.devops_kennel: Refreshing state... forgejo_team_member.devops_xboxbedrock: Refreshing state... forgejo_team_repository.devops_documentation: Refreshing state... forgejo_team_member.devops_thesuperrl: Refreshing state... forgejo_team_member.tartan-vote_leads_anish: Refreshing state... forgejo_team_member.quest_leads_xboxbedrock: Refreshing state... forgejo_team_member.tartan-vote_leads_maybe_yiyi: Refreshing state... forgejo_team_member.quest_leads_anish: Refreshing state... forgejo_team_repository.cmu-insta_insta: Refreshing state... forgejo_team_member.cmu-insta_etashjha: Refreshing state... forgejo_team_member.maps_leads_anish: Refreshing state... forgejo_team_member.maps_leads_siddharthr: Refreshing state... forgejo_team_member.cmugpt-agent_leads_krishsax: Refreshing state... forgejo_team_member.cmugpt-agent_leads_anish: Refreshing state... forgejo_team_member.dalmatian_leads_maybe_yiyi: Refreshing state... forgejo_team_member.dalmatian_leads_tentype: Refreshing state... forgejo_team_member.dalmatian_leads_anish: Refreshing state... forgejo_team_member.insta_leads_etashjha: Refreshing state... forgejo_team_member.discord-verify_leads_tentype: Refreshing state... forgejo_team_member.discord-verify_leads_maybe_yiyi: Refreshing state... forgejo_team_member.discord-verify_leads_anish: Refreshing state... forgejo_team_member.cbp_tentype: Refreshing state... forgejo_team_member.cbp_xboxbedrock: Refreshing state... forgejo_team_member.cbp_nikhilr: Refreshing state... forgejo_team_member.cbp_maybe_yiyi: Refreshing state... forgejo_team_member.cbp_liamwalk: Refreshing state... forgejo_team_repository.cbp_discord_verify: Refreshing state... forgejo_team_repository.cbp_bus_sign: Refreshing state... forgejo_team_member.cbp_anish: Refreshing state... forgejo_team_repository.cbp_dalmatian: Refreshing state... forgejo_team_repository.cbp_groupme_mirror: Refreshing state... forgejo_team_member.tartan-vote_maybe_yiyi: Refreshing state... forgejo_team_member.tartan-vote_xboxbedrock: Refreshing state... forgejo_team_member.tartan-vote_riztofu: Refreshing state... forgejo_team_repository.tartan-vote_tartan_vote: Refreshing state... forgejo_team_member.tartan-vote_anish: Refreshing state... forgejo_team_member.tartan-vote_tentype: Refreshing state... forgejo_team_member.cmu-maps_anish: Refreshing state... forgejo_team_repository.cmu-maps_maps: Refreshing state... forgejo_team_member.cmu-maps_siddharthr: Refreshing state... forgejo_team_member.cal_leads_jalenluorion: Refreshing state... forgejo_team_member.cal_leads_anish: Refreshing state... forgejo_team_member.terrier_leads_anish: Refreshing state... forgejo_team_member.terrier_leads_kritdass: Refreshing state... forgejo_team_member.cmugpt-surface_leads_krishsax: Refreshing state... forgejo_team_member.cmugpt-surface_leads_anish: Refreshing state... forgejo_team_member.components_leads_anish: Refreshing state... forgejo_team_repository.lost-and-found_lost_and_found_v2: Refreshing state... forgejo_team_member.lost-and-found_anish: Refreshing state... forgejo_team_member.housing_leads_emmawong: Refreshing state... forgejo_team_member.housing_leads_anish: Refreshing state... forgejo_team_member.housing_leads_tentype: Refreshing state... forgejo_team_member.housing_leads_nikhilr: Refreshing state... forgejo_team_member.housing_leads_gostmeaper: Refreshing state... forgejo_team_member.link-shortener_leads_anish: Refreshing state... forgejo_team_member.link-shortener_leads_thesuperrl: Refreshing state... forgejo_team_member.groupme-mirror_leads_anish: Refreshing state... forgejo_team_member.groupme-mirror_leads_tentype: Refreshing state... forgejo_team_member.groupme-mirror_leads_maybe_yiyi: Refreshing state... forgejo_team_member.cmugpt-mcp-server_leads_anish: Refreshing state... forgejo_team_member.cmugpt-mcp-server_leads_krishsax: Refreshing state... github_team_membership.cbp_liamwalk: Refreshing state... [id=17170099:pandaaured] github_team_membership.cbp_anish: Refreshing state... [id=17170099:ap-1] github_team_membership.cbp_maybe_yiyi: Refreshing state... [id=17170099:maybe-yiyi] github_team_membership.cbp_xboxbedrock: Refreshing state... [id=17170099:xboxbedrock] github_team_membership.cbp_tentype: Refreshing state... [id=17170099:tentype] github_team_membership.cbp_nikhilr: Refreshing state... [id=17170099:ecstaticpilot] github_actions_repository_permissions.devops_collie_actions: Refreshing state... [id=collie] github_team_membership.cmu-courses_anish: Refreshing state... [id=3643907:ap-1] github_team_membership.cmu-courses_etashjha: Refreshing state... [id=3643907:etashj] github_team_membership.cmu-courses_dkou: Refreshing state... [id=3643907:sharkree] github_team_membership.cmu-courses_gabrielx: Refreshing state... [id=3643907:gabriel-xu] github_team_membership.occ_xboxbedrock: Refreshing state... [id=19240310:xboxbedrock] github_team_membership.occ_dcrodrig: Refreshing state... [id=19240310:newhacker1746] github_team_membership.occ_acchs: Refreshing state... [id=19240310:acchs-at-cmu] github_team_membership.occ_anish: Refreshing state... [id=19240310:ap-1] github_team_membership.cmu-eats_arom1a: Refreshing state... [id=16807600:arom1a] github_team_membership.cmu-insta_etashjha: Refreshing state... [id=19240311:etashj] github_actions_repository_permissions.cmu-eats_eats_actions: Refreshing state... [id=eats] github_actions_repository_permissions.devops_kennel_actions: Refreshing state... [id=kennel] github_repository_deploy_key.devops_kennel_mirror_deploy_key: Refreshing state... [id=kennel:159665328] github_actions_repository_permissions.cmu-insta_insta_actions: Refreshing state... [id=insta] github_actions_repository_permissions.cmu-eats_dining_api_actions: Refreshing state... [id=dining-api] github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Refreshing state... [id=dining-api:161772815] github_repository_deploy_key.slai_cmugpt_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-surface:159665348] github_actions_repository_permissions.slai_cmugpt_surface_actions: Refreshing state... [id=cmugpt-surface] github_actions_repository_permissions.cbp_groupme_mirror_actions: Refreshing state... [id=groupme-mirror] github_repository_deploy_key.cbp_groupme_mirror_mirror_deploy_key: Refreshing state... [id=groupme-mirror:159665366] github_actions_repository_permissions.cmu-cal_cal_actions: Refreshing state... [id=cal] github_repository_deploy_key.cmu-cal_cal_mirror_deploy_key: Refreshing state... [id=cal:159665344] github_actions_repository_permissions.slai_cmugpt_sms_surface_actions: Refreshing state... [id=cmugpt-sms-surface] github_repository_deploy_key.slai_cmugpt_sms_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-sms-surface:159665365] github_actions_repository_permissions.devops_link_shortener_actions: Refreshing state... [id=link-shortener] github_team_membership.terrier_amzoeee: Refreshing state... [id=12028577:amzoeee] github_team_membership.terrier_kothurvidur: Refreshing state... [id=12028577:kothurvidur] github_team_membership.terrier_kritdass: Refreshing state... [id=12028577:kritdass] github_team_membership.terrier_dkou: Refreshing state... [id=12028577:sharkree] github_team_membership.terrier_etashjha: Refreshing state... [id=12028577:etashj] github_team_membership.terrier_anish: Refreshing state... [id=12028577:ap-1] github_team_membership.devops_xboxbedrock: Refreshing state... [id=14706218:xboxbedrock] github_team_membership.devops_thesuperrl: Refreshing state... [id=14706218:thesuperrl] github_team_membership.devops_anish: Refreshing state... [id=14706218:ap-1] github_actions_repository_permissions.occ_occ_actions: Refreshing state... [id=occ] github_actions_repository_permissions.devops_infrastructure_actions: Refreshing state... [id=infrastructure] github_repository_deploy_key.devops_infrastructure_mirror_deploy_key: Refreshing state... [id=infrastructure:159665343] github_actions_repository_permissions.uia_components_actions: Refreshing state... [id=components] github_actions_repository_permissions.lost-and-found_lost_and_found_v2_actions: Refreshing state... [id=lost-and-found-v2] github_repository_deploy_key.lost-and-found_lost_and_found_v2_mirror_deploy_key: Refreshing state... [id=lost-and-found-v2:159665337] github_actions_repository_permissions.cbp_discord_verify_actions: Refreshing state... [id=discord-verify] github_repository_deploy_key.cbp_discord_verify_mirror_deploy_key: Refreshing state... [id=discord-verify:159665340] github_team_membership.cmu-cal_anish: Refreshing state... [id=15669112:ap-1] github_team_membership.cmu-cal_jalenluorion: Refreshing state... [id=15669112:jalenluorion] github_actions_repository_permissions.cbp_dalmatian_actions: Refreshing state... [id=dalmatian] github_team_membership.uia_anish: Refreshing state... [id=17529763:ap-1] github_team_membership.slai_xboxbedrock: Refreshing state... [id=17571328:xboxbedrock] github_team_membership.slai_jiyas: Refreshing state... [id=17571328:jiya2301] github_team_membership.slai_rzhu3: Refreshing state... [id=17571328:richard-31415] github_team_membership.slai_qianxuege: Refreshing state... [id=17571328:qianxuege] github_team_membership.slai_krishsax: Refreshing state... [id=17571328:krishsaxena] github_team_membership.slai_jc_965: Refreshing state... [id=17571328:jc-965] github_team_membership.slai_anish: Refreshing state... [id=17571328:ap-1] github_actions_repository_permissions.slai_cmugpt_agent_actions: Refreshing state... [id=cmugpt-agent] github_repository_deploy_key.slai_cmugpt_agent_mirror_deploy_key: Refreshing state... [id=cmugpt-agent:159665360] github_actions_repository_permissions.quest_quest_actions: Refreshing state... [id=quest] github_repository_deploy_key.quest_quest_mirror_deploy_key: Refreshing state... [id=quest:159665323] github_actions_repository_permissions.uia_scottylabs_org_actions: Refreshing state... [id=scottylabs-org] github_repository_deploy_key.uia_scottylabs_org_mirror_deploy_key: Refreshing state... [id=scottylabs-org:159665361] github_actions_repository_permissions.cmu-housing_housing_actions: Refreshing state... [id=housing] github_repository_deploy_key.cmu-housing_housing_mirror_deploy_key: Refreshing state... [id=housing:159665369] github_actions_repository_permissions.cmu-courses_courses_actions: Refreshing state... [id=courses] github_repository_deploy_key.cmu-courses_courses_mirror_deploy_key: Refreshing state... [id=courses:159665327] github_actions_repository_permissions.terrier_terrier_actions: Refreshing state... [id=terrier] github_repository_deploy_key.terrier_terrier_mirror_deploy_key: Refreshing state... [id=terrier:159665339] github_actions_repository_permissions.devops_governance_actions: Refreshing state... [id=governance] github_repository_deploy_key.devops_governance_mirror_deploy_key: Refreshing state... [id=governance:159665346] github_team_membership.lost-and-found_anish: Refreshing state... [id=4088608:ap-1] github_team_membership.cmu-maps_siddharthr: Refreshing state... [id=10872360:overthrowing] github_team_membership.cmu-maps_anish: Refreshing state... [id=10872360:ap-1] github_repository_deploy_key.occ_occ_mirror_deploy_key: Refreshing state... [id=occ:161720242] github_repository_deploy_key.cbp_dalmatian_mirror_deploy_key: Refreshing state... [id=dalmatian:159665367] github_repository_deploy_key.devops_link_shortener_mirror_deploy_key: Refreshing state... [id=link-shortener:159665345] github_repository_deploy_key.devops_documentation_mirror_deploy_key: Refreshing state... [id=documentation:159665362] github_actions_repository_permissions.devops_documentation_actions: Refreshing state... [id=documentation] github_repository_deploy_key.devops_collie_mirror_deploy_key: Refreshing state... [id=collie:159665332] github_repository_deploy_key.uia_components_mirror_deploy_key: Refreshing state... [id=components:159665353] github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Refreshing state... [id=eats:161775995] github_repository_deploy_key.cmu-insta_insta_mirror_deploy_key: Refreshing state... [id=insta:161720245] github_repository_deploy_key.tartan-vote_tartan_vote_mirror_deploy_key: Refreshing state... [id=tartan-vote:159665350] github_actions_repository_permissions.tartan-vote_tartan_vote_actions: Refreshing state... [id=tartan-vote] forgejo_team_repository.terrier_leads_terrier: Refreshing state... forgejo_team_repository.housing_leads_housing: Refreshing state... forgejo_team_repository.scottylabs-org_leads_scottylabs_org: Refreshing state... forgejo_team_repository.components_leads_components: Refreshing state... forgejo_team_repository.cal_leads_cal: Refreshing state... forgejo_team_repository.occ-main_leads_occ: Refreshing state... forgejo_team_repository.quest_leads_quest: Refreshing state... forgejo_team_repository.cmugpt-surface_leads_cmugpt_surface: Refreshing state... forgejo_team_repository.cmugpt-agent_leads_cmugpt_agent: Refreshing state... forgejo_team_repository.cmugpt-sms-surface_leads_cmugpt_sms_surface: Refreshing state... forgejo_team_repository.cmugpt-mcp-server_leads_mcp_server: Refreshing state... forgejo_team_repository.cmu-courses_leads_internet_archive: Refreshing state... forgejo_team_repository.courses_leads_courses: Refreshing state... forgejo_team_repository.devops_leads_infrastructure: Refreshing state... forgejo_team_repository.link-shortener_leads_link_shortener: Refreshing state... forgejo_team_repository.devops_leads_governance: Refreshing state... forgejo_team_repository.devops_leads_collie: Refreshing state... forgejo_team_repository.devops_leads_kennel: Refreshing state... forgejo_team_repository.documentation_leads_documentation: Refreshing state... forgejo_team_repository.insta_leads_insta: Refreshing state... forgejo_team_repository.discord-verify_leads_discord_verify: Refreshing state... forgejo_team_repository.bus-sign_leads_bus_sign: Refreshing state... forgejo_team_repository.dalmatian_leads_dalmatian: Refreshing state... forgejo_team_repository.groupme-mirror_leads_groupme_mirror: Refreshing state... forgejo_team_repository.tartan-vote_leads_tartan_vote: Refreshing state... github_repository_deploy_key.cmu-courses_internet_archive_mirror_deploy_key: Refreshing state... [id=internet-archive:159665356] github_actions_repository_permissions.cmu-courses_internet_archive_actions: Refreshing state... [id=internet-archive] forgejo_team_repository.maps_leads_maps: Refreshing state... forgejo_team_repository.lost-and-found-v2_leads_lost_and_found_v2: Refreshing state... github_team_membership.tartan-vote_maybe_yiyi: Refreshing state... [id=17701833:maybe-yiyi] github_team_membership.tartan-vote_xboxbedrock: Refreshing state... [id=17701833:xboxbedrock] github_team_membership.tartan-vote_tentype: Refreshing state... [id=17701833:tentype] github_team_membership.tartan-vote_riztofu: Refreshing state... [id=17701833:riztofu] github_team_membership.tartan-vote_anish: Refreshing state... [id=17701833:ap-1] github_team_membership.cmu-housing_amzoeee: Refreshing state... [id=17937919:amzoeee] github_team_membership.cmu-housing_xboxbedrock: Refreshing state... [id=17937919:xboxbedrock] github_team_membership.cmu-housing_nikhilr: Refreshing state... [id=17937919:ecstaticpilot] github_team_membership.cmu-housing_rhsen: Refreshing state... [id=17937919:rohansen08] github_team_membership.cmu-housing_jessica_lr_song: Refreshing state... [id=17937919:jessica-lr-song] github_team_membership.cmu-housing_tentype: Refreshing state... [id=17937919:tentype] github_team_membership.cmu-housing_stringatlas: Refreshing state... [id=17937919:stringatlas] github_team_membership.cmu-housing_anish: Refreshing state... [id=17937919:ap-1] github_team_membership.cmu-housing_emmawong: Refreshing state... [id=17937919:emw868] github_team_membership.cmu-housing_manrajs: Refreshing state... [id=17937919:manraj10] github_team_membership.cmu-housing_gostmeaper: Refreshing state... [id=17937919:gostmeaper] github_team_membership.cmu-housing_kluoxx: Refreshing state... [id=17937919:luokaden888] github_team_membership.cmu-housing_zoexue: Refreshing state... [id=17937919:o-zox-o] github_team_membership.quest_xboxbedrock: Refreshing state... [id=17330197:xboxbedrock] github_team_membership.quest_brightz: Refreshing state... [id=17330197:bright-zheng] github_team_membership.quest_jw8: Refreshing state... [id=17330197:jeffreyyijiwang] github_team_membership.quest_kechezon: Refreshing state... [id=17330197:kechezon] github_team_membership.quest_anish: Refreshing state... [id=17330197:ap-1] github_actions_repository_permissions.cbp_bus_sign_actions: Refreshing state... [id=bus-sign] github_repository_deploy_key.cbp_bus_sign_mirror_deploy_key: Refreshing state... [id=bus-sign:159665342] github_actions_repository_permissions.slai_mcp_server_actions: Refreshing state... [id=mcp-server] github_repository_deploy_key.slai_mcp_server_mirror_deploy_key: Refreshing state... [id=mcp-server:159665358] github_repository_deploy_key.cmu-maps_maps_mirror_deploy_key: Refreshing state... [id=maps:159665325] github_actions_repository_permissions.cmu-maps_maps_actions: Refreshing state... [id=maps] github_repository_ruleset.devops_kennel_ruleset: Refreshing state... [id=13193453] github_repository_ruleset.cmu-eats_dining_api_ruleset: Refreshing state... [id=21879626] github_repository_ruleset.slai_cmugpt_surface_ruleset: Refreshing state... [id=16476662] github_repository_ruleset.cbp_groupme_mirror_ruleset: Refreshing state... [id=12503296] github_repository_ruleset.cmu-cal_cal_ruleset: Refreshing state... [id=18919281] github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Refreshing state... [id=17653622] github_repository_ruleset.devops_infrastructure_ruleset: Refreshing state... [id=11275118] github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Refreshing state... [id=18920618] github_repository_ruleset.cbp_discord_verify_ruleset: Refreshing state... [id=11449154] github_repository_ruleset.slai_cmugpt_agent_ruleset: Refreshing state... [id=16476661] github_repository_ruleset.quest_quest_ruleset: Refreshing state... [id=15750052] github_repository_ruleset.uia_scottylabs_org_ruleset: Refreshing state... [id=18112214] github_repository_ruleset.cmu-housing_housing_ruleset: Refreshing state... [id=17421604] github_repository_ruleset.cmu-courses_courses_ruleset: Refreshing state... [id=15855161] github_repository_ruleset.terrier_terrier_ruleset: Refreshing state... [id=12671003] github_repository_ruleset.devops_governance_ruleset: Refreshing state... [id=14488792] github_repository_ruleset.occ_occ_ruleset: Refreshing state... [id=21841634] github_repository_ruleset.cbp_dalmatian_ruleset: Refreshing state... [id=17238915] github_repository_ruleset.devops_link_shortener_ruleset: Refreshing state... [id=17911647] github_repository_ruleset.devops_documentation_ruleset: Refreshing state... [id=17079368] github_repository_ruleset.devops_collie_ruleset: Refreshing state... [id=18865204] github_repository_ruleset.uia_components_ruleset: Refreshing state... [id=16366085] github_repository_ruleset.cmu-eats_eats_ruleset: Refreshing state... [id=21330218] github_repository_ruleset.cmu-insta_insta_ruleset: Refreshing state... [id=21841636] github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Refreshing state... [id=16848966] github_repository_ruleset.cmu-courses_internet_archive_ruleset: Refreshing state... [id=11988074] github_repository_ruleset.cbp_bus_sign_ruleset: Refreshing state... [id=15236550] github_repository_ruleset.slai_mcp_server_ruleset: Refreshing state... [id=17082154] github_repository_ruleset.cmu-maps_maps_ruleset: Refreshing state... [id=18864961] Note: Objects have changed outside of OpenTofu OpenTofu detected the following changes made outside of OpenTofu since the last "tofu apply" which may have affected this plan: # restapi_object.cmu-eats_dining_api_push_mirror has been deleted - resource "restapi_object" "cmu-eats_dining_api_push_mirror" { - api_data = { - "public_key" = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIIKyjdH3wH8J3aSgeZMLB9/uUGzyYltQsk5Cl1Q8MXqF" # (9 unchanged elements hidden) } -> null id = "remote_mirror_a4MsQ5rO1_Z" # (5 unchanged attributes hidden) } # restapi_object.cmu-eats_eats_push_mirror has been deleted - resource "restapi_object" "cmu-eats_eats_push_mirror" { - api_data = { - "public_key" = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFsm79Zlyi0JKH4y/BdgCCoLyrC4/18yF3owfsqOt53/" # (9 unchanged elements hidden) } -> null id = "remote_mirror_sjYwMA6NPgh" # (5 unchanged attributes hidden) } Unless you have made equivalent changes to your configuration, or ignored the relevant attributes using ignore_changes, the following plan may include actions to undo or respond to these changes. ───────────────────────────────────────────────────────────────────────────── OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ~ update in-place (current -> planned) -/+ destroy and then create replacement OpenTofu planned the following actions, but then encountered a problem: # forgejo_repository.cmu-eats_dining_api will be updated in-place ~ resource "forgejo_repository" "cmu-eats_dining_api" { ~ clone_url = "https://git.cmu.dev/ScottyLabs/dining-api.git" -> (known after apply) ~ created_at = "2026-09-18T23:16:27Z" -> (known after apply) + description = "Dining API for CMU dining locations and availability" ~ empty = false -> (known after apply) + external_tracker = (known after apply) + external_wiki = (known after apply) ~ forks_count = 0 -> (known after apply) ~ full_name = "ScottyLabs/dining-api" -> (known after apply) ~ html_url = "https://git.cmu.dev/ScottyLabs/dining-api" -> (known after apply) id = 97 ~ internal_tracker = { ~ allow_only_contributors_to_track_time = true -> (known after apply) ~ enable_issue_dependencies = true -> (known after apply) ~ enable_time_tracker = true -> (known after apply) } -> (known after apply) + mirror_interval = (known after apply) ~ mirror_updated = "0001-01-01T00:00:00Z" -> (known after apply) name = "dining-api" ~ open_issues_count = 0 -> (known after apply) ~ open_pr_counter = 5 -> (known after apply) ~ permissions = { ~ admin = true -> (known after apply) ~ pull = true -> (known after apply) ~ push = true -> (known after apply) } -> (known after apply) ~ release_counter = 0 -> (known after apply) ~ size = 38816 -> (known after apply) ~ ssh_url = "ssh://forgejo@git.cmu.dev/ScottyLabs/dining-api.git" -> (known after apply) ~ stars_count = 0 -> (known after apply) ~ updated_at = "2026-09-18T23:22:22Z" -> (known after apply) ~ watchers_count = 3 -> (known after apply) ~ website = "https://git.cmu.dev/ScottyLabs/dining-api" -> "https://api.cmueats.com" # (30 unchanged attributes hidden) } # forgejo_repository.cmu-eats_eats will be updated in-place ~ resource "forgejo_repository" "cmu-eats_eats" { ~ clone_url = "https://git.cmu.dev/ScottyLabs/eats.git" -> (known after apply) ~ created_at = "2026-09-18T23:15:39Z" -> (known after apply) ~ description = "Vite web app for CMU dining location info" -> "CMU dining location info" ~ empty = false -> (known after apply) + external_tracker = (known after apply) + external_wiki = (known after apply) ~ forks_count = 0 -> (known after apply) ~ full_name = "ScottyLabs/eats" -> (known after apply) ~ html_url = "https://git.cmu.dev/ScottyLabs/eats" -> (known after apply) id = 96 ~ internal_tracker = { ~ allow_only_contributors_to_track_time = true -> (known after apply) ~ enable_issue_dependencies = true -> (known after apply) ~ enable_time_tracker = true -> (known after apply) } -> (known after apply) + mirror_interval = (known after apply) ~ mirror_updated = "0001-01-01T00:00:00Z" -> (known after apply) name = "eats" ~ open_issues_count = 48 -> (known after apply) ~ open_pr_counter = 4 -> (known after apply) ~ permissions = { ~ admin = true -> (known after apply) ~ pull = true -> (known after apply) ~ push = true -> (known after apply) } -> (known after apply) ~ release_counter = 0 -> (known after apply) ~ size = 95496 -> (known after apply) ~ ssh_url = "ssh://forgejo@git.cmu.dev/ScottyLabs/eats.git" -> (known after apply) ~ stars_count = 0 -> (known after apply) ~ updated_at = "2026-09-18T23:35:12Z" -> (known after apply) ~ watchers_count = 3 -> (known after apply) # (31 unchanged attributes hidden) } # forgejo_repository.cmu-study_study will be created + resource "forgejo_repository" "cmu-study_study" { + allow_manual_merge = false + allow_merge_commits = true + allow_rebase = true + allow_rebase_explicit = true + allow_squash_merge = true + archive_on_destroy = false + archived = false + auto_init = true + autodetect_manual_merge = false + avatar_url = (known after apply) + clone_url = (known after apply) + created_at = (known after apply) + default_branch = "main" + default_merge_style = "merge" + description = "A platform for creating and joining study groups" + empty = (known after apply) + external_tracker = (known after apply) + external_wiki = (known after apply) + fork = (known after apply) + forks_count = (known after apply) + full_name = (known after apply) + has_actions = true + has_issues = true + has_packages = true + has_projects = true + has_pull_requests = true + has_releases = true + has_wiki = true + html_url = (known after apply) + id = (known after apply) + ignore_whitespace_conflicts = false + internal = (known after apply) + internal_tracker = (known after apply) + labels = false + lfs = false + milestones = false + mirror = false + mirror_interval = (known after apply) + mirror_updated = (known after apply) + name = "study" + open_issues_count = (known after apply) + open_pr_counter = (known after apply) + owner = "ScottyLabs" + parent_id = (known after apply) + permissions = (known after apply) + private = false + release_counter = (known after apply) + size = (known after apply) + ssh_url = (known after apply) + stars_count = (known after apply) + template = false + trust_model = "default" + updated_at = (known after apply) + watchers_count = (known after apply) + website = "https://cmustudy.com" } # forgejo_team.cmu-study will be created + resource "forgejo_team" "cmu-study" { + can_create_org_repo = (known after apply) + description = "CMU Study" + id = (known after apply) + includes_all_repositories = (known after apply) + name = "cmu-study" + organization = (known after apply) + organization_id = 3 + permission = "write" + units_map = { + "repo.code" = "write" + "repo.issues" = "write" + "repo.pulls" = "write" } } # forgejo_team.study_leads will be created + resource "forgejo_team" "study_leads" { + can_create_org_repo = false + description = "CMU Study leads" + id = (known after apply) + includes_all_repositories = false + name = "study-leads" + organization = (known after apply) + organization_id = 3 + permission = "admin" + units_map = { + "repo.actions" = "admin" + "repo.code" = "admin" + "repo.ext_issues" = "admin" + "repo.ext_wiki" = "admin" + "repo.issues" = "admin" + "repo.packages" = "admin" + "repo.projects" = "admin" + "repo.pulls" = "admin" + "repo.releases" = "admin" + "repo.wiki" = "admin" } } # forgejo_team_member.cmu-study_annadavi will be created + resource "forgejo_team_member" "cmu-study_annadavi" { + team_id = (known after apply) + user = "annadavi" } # forgejo_team_member.study_leads_annadavi will be created + resource "forgejo_team_member" "study_leads_annadavi" { + team_id = (known after apply) + user = "annadavi" } # forgejo_team_repository.cmu-study_study will be created + resource "forgejo_team_repository" "cmu-study_study" { + owner = "ScottyLabs" + repository = "study" + team_id = (known after apply) } # forgejo_team_repository.study_leads_study will be created + resource "forgejo_team_repository" "study_leads_study" { + owner = "ScottyLabs" + repository = "study" + team_id = (known after apply) } # github_actions_repository_permissions.cmu-study_study_actions will be created + resource "github_actions_repository_permissions" "cmu-study_study_actions" { + enabled = false + id = (known after apply) + repository = "study" + sha_pinning_required = (known after apply) } # github_repository.cmu-study_study will be created + resource "github_repository" "cmu-study_study" { + allow_auto_merge = false + allow_forking = (known after apply) + allow_merge_commit = true + allow_rebase_merge = true + allow_squash_merge = true + archived = false + default_branch = (known after apply) + delete_branch_on_merge = false + description = "READ-ONLY MIRROR" + etag = (known after apply) + fork = (known after apply) + full_name = (known after apply) + git_clone_url = (known after apply) + has_discussions = false + has_issues = false + has_projects = false + has_wiki = false + homepage_url = "https://git.cmu.dev/ScottyLabs/study" + html_url = (known after apply) + http_clone_url = (known after apply) + id = (known after apply) + ignore_vulnerability_alerts_during_read = false + merge_commit_message = "PR_TITLE" + merge_commit_title = "MERGE_MESSAGE" + name = "study" + node_id = (known after apply) + primary_language = (known after apply) + private = (known after apply) + repo_id = (known after apply) + source_owner = (known after apply) + source_repo = (known after apply) + squash_merge_commit_message = "COMMIT_MESSAGES" + squash_merge_commit_title = "COMMIT_OR_PR_TITLE" + ssh_clone_url = (known after apply) + svn_url = (known after apply) + topics = (known after apply) + visibility = "public" + vulnerability_alerts = (known after apply) + web_commit_signoff_required = (known after apply) + security_and_analysis (known after apply) } # github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key must be replaced -/+ resource "github_repository_deploy_key" "cmu-eats_dining_api_mirror_deploy_key" { ~ etag = "W/\"bf394a0fcc9e1231b4ee544598e6680359d71d177764d9f0c8cd22f5c7ec3ef4\"" -> (known after apply) ~ id = "dining-api:161772815" -> (known after apply) ~ key = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIIKyjdH3wH8J3aSgeZMLB9/uUGzyYltQsk5Cl1Q8MXqF" # forces replacement -> (known after apply) # forces replacement # (3 unchanged attributes hidden) } # github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key must be replaced -/+ resource "github_repository_deploy_key" "cmu-eats_eats_mirror_deploy_key" { ~ etag = "W/\"5cf445e3b52f1b2e59fe1cb32c5c2e0859e2598566a7aa75c3ebdbf4b01f6164\"" -> (known after apply) ~ id = "eats:161775995" -> (known after apply) ~ key = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFsm79Zlyi0JKH4y/BdgCCoLyrC4/18yF3owfsqOt53/" # forces replacement -> (known after apply) # forces replacement # (3 unchanged attributes hidden) } # github_repository_deploy_key.cmu-study_study_mirror_deploy_key will be created + resource "github_repository_deploy_key" "cmu-study_study_mirror_deploy_key" { + etag = (known after apply) + id = (known after apply) + key = (known after apply) + read_only = false + repository = "study" + title = "Forgejo mirror" } # github_repository_ruleset.cmu-eats_dining_api_ruleset will be updated in-place ~ resource "github_repository_ruleset" "cmu-eats_dining_api_ruleset" { id = "21879626" name = "Default" # (6 unchanged attributes hidden) ~ bypass_actors { ~ actor_id = 161772815 -> (known after apply) # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place ~ resource "github_repository_ruleset" "cmu-eats_eats_ruleset" { id = "21330218" name = "Default" # (6 unchanged attributes hidden) ~ bypass_actors { ~ actor_id = 161775995 -> (known after apply) # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.cmu-study_study_ruleset will be created + resource "github_repository_ruleset" "cmu-study_study_ruleset" { + enforcement = "active" + etag = (known after apply) + id = (known after apply) + name = "Default" + node_id = (known after apply) + repository = "study" + ruleset_id = (known after apply) + target = "branch" + bypass_actors { + actor_id = (known after apply) + actor_type = "DeployKey" + bypass_mode = "always" } + conditions { + ref_name { + exclude = [] + include = [ + "~ALL", ] } } + rules { + creation = true + deletion = true + non_fast_forward = true + update = true + update_allows_fetch_and_merge = false + pull_request { + allowed_merge_methods = (known after apply) + dismiss_stale_reviews_on_push = false + require_code_owner_review = false + require_last_push_approval = false + required_approving_review_count = 1 + required_review_thread_resolution = false } } } # github_team.cmu-study will be created + resource "github_team" "cmu-study" { + create_default_maintainer = false + etag = (known after apply) + id = (known after apply) + members_count = (known after apply) + name = "CMU Study" + node_id = (known after apply) + notification_setting = "notifications_enabled" + parent_team_read_id = (known after apply) + parent_team_read_slug = (known after apply) + privacy = "closed" + slug = (known after apply) } # github_team_membership.cmu-study_annadavi will be created + resource "github_team_membership" "cmu-study_annadavi" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = (known after apply) + username = "annadavid1" } # restapi_object.cmu-eats_dining_api_push_mirror will be created + resource "restapi_object" "cmu-eats_dining_api_push_mirror" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = jsonencode( { + interval = "8h0m0s" + remote_address = "git@github.com:ScottyLabs/dining-api.git" + sync_on_commit = true + use_ssh = true } ) + id = (known after apply) + id_attribute = "remote_name" + path = "/api/v1/repos/ScottyLabs/dining-api/push_mirrors" } # restapi_object.cmu-eats_eats_kennel_webhook will be created + resource "restapi_object" "cmu-eats_eats_kennel_webhook" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = (sensitive value) + id = (known after apply) + id_attribute = "id" + path = "/api/v1/repos/ScottyLabs/eats/hooks" + update_method = "PATCH" } # restapi_object.cmu-eats_eats_push_mirror will be created + resource "restapi_object" "cmu-eats_eats_push_mirror" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = jsonencode( { + interval = "8h0m0s" + remote_address = "git@github.com:ScottyLabs/eats.git" + sync_on_commit = true + use_ssh = true } ) + id = (known after apply) + id_attribute = "remote_name" + path = "/api/v1/repos/ScottyLabs/eats/push_mirrors" } # restapi_object.cmu-study_study_kennel_webhook will be created + resource "restapi_object" "cmu-study_study_kennel_webhook" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = (sensitive value) + id = (known after apply) + id_attribute = "id" + path = "/api/v1/repos/ScottyLabs/study/hooks" + update_method = "PATCH" } # restapi_object.cmu-study_study_push_mirror will be created + resource "restapi_object" "cmu-study_study_push_mirror" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = jsonencode( { + interval = "8h0m0s" + remote_address = "git@github.com:ScottyLabs/study.git" + sync_on_commit = true + use_ssh = true } ) + id = (known after apply) + id_attribute = "remote_name" + path = "/api/v1/repos/ScottyLabs/study/push_mirrors" } Plan: 20 to add, 4 to change, 2 to destroy. ╷ │ Error: Unable to read team repository │ │ with forgejo_team_repository.cmu-eats_dining_api, │ on forgejo_team_repos.tf.json line 102, in resource.forgejo_team_repository.cmu-eats_dining_api: │ 102: }, │ │ Repository 'ScottyLabs/dining-api' not found in team with ID 28 ╵ ╷ │ Error: Unable to read team repository │ │ with forgejo_team_repository.cmu-eats_eats, │ on forgejo_team_repos.tf.json line 111, in resource.forgejo_team_repository.cmu-eats_eats: │ 111: }, │ │ Repository 'ScottyLabs/eats' not found in team with ID 28 ╵ ``` </details>
anish force-pushed main from 8c41fd4672
Some checks failed
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 3 to add, 0 to change, 0 to destroy.
atlantis/plan: openbao Plan: 6 to add, 24 to change, 0 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: posthog Plan: 3 to add, 0 to change, 0 to destroy.
atlantis/plan: matrix_bridges Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: sentry Plan: 4 to add, 0 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 38 to add, 1 to change, 0 to destroy.
atlantis/plan: discord Plan: 3 to add, 0 to change, 0 to destroy.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: git Plan failed.
atlantis/plan 1/10 projects failed to plan.
to d53b47c18f
All checks were successful
atlantis/apply: google_play Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: openbao Apply succeeded.
atlantis/apply: posthog Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: sentry Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: matrix_bridges Apply succeeded.
atlantis/plan: git Plan: 7 to add, 1 to change, 0 to destroy.
atlantis/plan 10/10 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply 10/10 projects applied successfully.
atlantis/plan: posthog Plan: 3 to add, 0 to change, 0 to destroy.
atlantis/plan: matrix_bridges Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: slack Plan: 3 to add, 0 to change, 0 to destroy.
atlantis/plan: openbao Plan: 6 to add, 24 to change, 0 to destroy.
atlantis/plan: sentry Plan: 4 to add, 0 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 38 to add, 1 to change, 0 to destroy.
atlantis/plan: discord Plan: 3 to add, 0 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
2026-09-19 00:04:30 +00:00
Compare
Owner

atlantis state -p git rm forgejo_team_repository.cmu-eats_dining_api forgejo_team_repository.cmu-eats_eats

atlantis state -p git rm forgejo_team_repository.cmu-eats_dining_api forgejo_team_repository.cmu-eats_eats

Ran State rm for project: git dir: tofu/git workspace: default

State Error

cannot run "state": the default workspace at path tofu/git is currently locked for this pull request by "plan" for commit d53b47c.
Wait until the previous command is complete and try again
Ran State `rm` for project: `git` dir: `tofu/git` workspace: `default` **State Error** ``` cannot run "state": the default workspace at path tofu/git is currently locked for this pull request by "plan" for commit d53b47c. Wait until the previous command is complete and try again ``` - Commit: [d53b47c18fc0c854359a03dfed26fae1b3f511a8](https://git.cmu.dev/ScottyLabs/governance/pulls/215/commits/d53b47c18fc0c854359a03dfed26fae1b3f511a8) - Blocking job: [View Atlantis execution](https://atlantis.scottylabs.org/jobs/b3831852-9532-4306-8a9a-3f28af3989cc)

Ran Plan for 10 projects:

  1. project: posthog dir: tofu/posthog workspace: default
  2. project: matrix_bridges dir: tofu/matrix_bridges workspace: default
  3. project: slack dir: tofu/slack workspace: default
  4. project: openbao dir: tofu/openbao workspace: default
  5. project: google_play dir: tofu/google_play workspace: default
  6. project: sentry dir: tofu/sentry workspace: default
  7. project: google_groups dir: tofu/google_groups workspace: default
  8. project: keycloak dir: tofu/keycloak workspace: default
  9. project: discord dir: tofu/discord workspace: default
  10. project: git dir: tofu/git workspace: default

1. project: posthog dir: tofu/posthog workspace: default

Show Output
Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.10s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # posthog_project.cmu-study_study will be created
+ resource "posthog_project" "cmu-study_study" {
+       api_token       = (sensitive value)
+       id              = (known after apply)
+       name            = "study"
+       organization_id = (known after apply)
+       timezone        = (known after apply)
    }

  # vault_kv_secret_v2.cmu-study_study_posthog_host will be created
+ resource "vault_kv_secret_v2" "cmu-study_study_posthog_host" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/POSTHOG_HOST"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.cmu-study_study_posthog_key will be created
+ resource "vault_kv_secret_v2" "cmu-study_study_posthog_key" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/POSTHOG_KEY"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

Plan: 3 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p posthog
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p posthog
    

Plan: 3 to add, 0 to change, 0 to destroy.


2. project: matrix_bridges dir: tofu/matrix_bridges workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.25s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_bridge_login_cmu_study will be created
+ resource "null_resource" "slack_bridge_login_cmu_study" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C09KN6M3P53"
        }
    }

  # synapse_mautrix_slack_link.cmu_study will be created
+ resource "synapse_mautrix_slack_link" "cmu_study" {
+       discord_channel_id       = "1512270354943578132"
+       discord_relay_configured = (known after apply)
+       id                       = (known after apply)
+       room_alias               = (known after apply)
+       room_id                  = (known after apply)
+       slack_bridged            = (known after apply)
+       slack_channel_id         = "C09KN6M3P53"
+       team_name                = "CMU Study"
+       team_slug                = "cmu-study"
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p matrix_bridges
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p matrix_bridges
    

Plan: 2 to add, 0 to change, 0 to destroy.


3. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.43s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-study_ch0_annadavi will be created
+ resource "null_resource" "slack_cmu-study_ch0_annadavi" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C09KN6M3P53"
          + "user"    = "U07M24AKRGV"
        }
    }

  # null_resource.slack_hub_annadavi will be created
+ resource "null_resource" "slack_hub_annadavi" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U07M24AKRGV"
        }
    }

  # null_resource.slack_leads_annadavi will be created
+ resource "null_resource" "slack_leads_annadavi" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0AN1M0HDV0"
          + "user"    = "U07M24AKRGV"
        }
    }

Plan: 3 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 3 to add, 0 to change, 0 to destroy.


4. project: openbao dir: tofu/openbao workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.25s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # vault_identity_group.bus-sign_admins will be updated in-place
! resource "vault_identity_group" "bus-sign_admins" {
        id                         = "1b72d8b5-8511-fe92-c98c-2ddb8cbf17a1"
        name                       = "bus-sign-admins"
!       policies                   = [
-           "bus-sign-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.cal_admins will be updated in-place
! resource "vault_identity_group" "cal_admins" {
        id                         = "da96ea82-8537-f0a1-0d42-a7ab1cc3c006"
        name                       = "cal-admins"
!       policies                   = [
-           "cal-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.changelog_admins will be updated in-place
! resource "vault_identity_group" "changelog_admins" {
        id                         = "d73c0ef4-a7d2-d171-ebd1-1c15cee174c8"
        name                       = "changelog-admins"
!       policies                   = [
-           "changelog-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.cmugpt-agent_admins will be updated in-place
! resource "vault_identity_group" "cmugpt-agent_admins" {
        id                         = "cb081cd2-a3cd-81e1-8c20-0914cac5c213"
        name                       = "cmugpt-agent-admins"
!       policies                   = [
-           "cmugpt-agent-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.cmugpt-mcp-server_admins will be updated in-place
! resource "vault_identity_group" "cmugpt-mcp-server_admins" {
        id                         = "a4717a6f-015f-9fae-c6a2-76657b8a7464"
        name                       = "cmugpt-mcp-server-admins"
!       policies                   = [
-           "cmugpt-mcp-server-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.cmugpt-sms-surface_admins will be updated in-place
! resource "vault_identity_group" "cmugpt-sms-surface_admins" {
        id                         = "780f9498-dee7-8cac-9bbc-e1feab5ea2fb"
        name                       = "cmugpt-sms-surface-admins"
!       policies                   = [
-           "cmugpt-sms-surface-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.cmugpt-surface_admins will be updated in-place
! resource "vault_identity_group" "cmugpt-surface_admins" {
        id                         = "3ae18ef3-f370-f163-767a-accfb092ea0d"
        name                       = "cmugpt-surface-admins"
!       policies                   = [
-           "cmugpt-surface-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.components_admins will be updated in-place
! resource "vault_identity_group" "components_admins" {
        id                         = "d2971366-64a4-984e-1e4e-dad3cd07c841"
        name                       = "components-admins"
!       policies                   = [
-           "components-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.courses_admins will be updated in-place
! resource "vault_identity_group" "courses_admins" {
        id                         = "d090ae17-1992-86a5-1f8b-092c937092e8"
        name                       = "courses-admins"
!       policies                   = [
-           "courses-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.dalmatian_admins will be updated in-place
! resource "vault_identity_group" "dalmatian_admins" {
        id                         = "b1d88b18-8b44-c2ea-9220-fc53ee7e3c7e"
        name                       = "dalmatian-admins"
!       policies                   = [
-           "dalmatian-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.discord-verify_admins will be updated in-place
! resource "vault_identity_group" "discord-verify_admins" {
        id                         = "c3e29b3b-0ce5-61a3-3a69-e950f297fa78"
        name                       = "discord-verify-admins"
!       policies                   = [
-           "discord-verify-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.documentation_admins will be updated in-place
! resource "vault_identity_group" "documentation_admins" {
        id                         = "35c5df83-0e17-31fc-7b82-3365d5526dd5"
        name                       = "documentation-admins"
!       policies                   = [
-           "documentation-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.eats_admins will be updated in-place
! resource "vault_identity_group" "eats_admins" {
        id                         = "05959815-c896-9703-32eb-7ba84604303c"
        name                       = "eats-admins"
!       policies                   = [
-           "eats-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.groupme-mirror_admins will be updated in-place
! resource "vault_identity_group" "groupme-mirror_admins" {
        id                         = "e264aabf-8766-67fd-f0bf-860a240ee2c2"
        name                       = "groupme-mirror-admins"
!       policies                   = [
-           "groupme-mirror-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.housing_admins will be updated in-place
! resource "vault_identity_group" "housing_admins" {
        id                         = "7c688f49-59f1-1593-1e19-16f70b86eae2"
        name                       = "housing-admins"
!       policies                   = [
-           "housing-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.insta_admins will be updated in-place
! resource "vault_identity_group" "insta_admins" {
        id                         = "be24e607-e8d5-9fa3-a11a-9a8ca05ede04"
        name                       = "insta-admins"
!       policies                   = [
-           "insta-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.link-shortener_admins will be updated in-place
! resource "vault_identity_group" "link-shortener_admins" {
        id                         = "d3fa8791-4bd4-db20-68b6-79db5ac36a25"
        name                       = "link-shortener-admins"
!       policies                   = [
-           "link-shortener-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.lost-and-found-v2_admins will be updated in-place
! resource "vault_identity_group" "lost-and-found-v2_admins" {
        id                         = "db8b5072-bb86-c9d0-e93f-951c2901bd15"
        name                       = "lost-and-found-v2-admins"
!       policies                   = [
-           "lost-and-found-v2-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.maps_admins will be updated in-place
! resource "vault_identity_group" "maps_admins" {
        id                         = "6b2bd85e-e311-ae83-8a07-30100f0ac12f"
        name                       = "maps-admins"
!       policies                   = [
-           "maps-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.occ-main_admins will be updated in-place
! resource "vault_identity_group" "occ-main_admins" {
        id                         = "4d40db70-1bac-505e-ec8d-5a9d5cf15ac7"
        name                       = "occ-main-admins"
!       policies                   = [
-           "occ-main-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.quest_admins will be updated in-place
! resource "vault_identity_group" "quest_admins" {
        id                         = "c03f2eab-1701-f315-fc17-39814ddcb34b"
        name                       = "quest-admins"
!       policies                   = [
-           "quest-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.scottylabs-org_admins will be updated in-place
! resource "vault_identity_group" "scottylabs-org_admins" {
        id                         = "0dbd914d-b96c-de7b-1ec5-a0ef330b20e0"
        name                       = "scottylabs-org-admins"
!       policies                   = [
-           "scottylabs-org-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.study_admins will be created
+ resource "vault_identity_group" "study_admins" {
+       external_member_entity_ids = false
+       external_member_group_ids  = false
+       external_policies          = false
+       id                         = (known after apply)
+       name                       = "study-admins"
+       policies                   = [
+           "study-prod",
        ]
+       type                       = "external"
    }

  # vault_identity_group.study_members will be created
+ resource "vault_identity_group" "study_members" {
+       external_member_entity_ids = false
+       external_member_group_ids  = false
+       external_policies          = false
+       id                         = (known after apply)
+       name                       = "study-members"
+       policies                   = [
+           "study-dev",
        ]
+       type                       = "external"
    }

  # vault_identity_group.tartan-vote_admins will be updated in-place
! resource "vault_identity_group" "tartan-vote_admins" {
        id                         = "ff77ad01-20b1-c320-37bd-6185b3f98d80"
        name                       = "tartan-vote-admins"
!       policies                   = [
-           "tartan-vote-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group.terrier_admins will be updated in-place
! resource "vault_identity_group" "terrier_admins" {
        id                         = "956692a1-84f6-c95c-6d40-2382ead13476"
        name                       = "terrier-admins"
!       policies                   = [
-           "terrier-dev",
            # (1 unchanged element hidden)
        ]
        # (7 unchanged attributes hidden)
    }

  # vault_identity_group_alias.study_admins will be created
+ resource "vault_identity_group_alias" "study_admins" {
+       canonical_id   = (known after apply)
+       id             = (known after apply)
+       mount_accessor = "auth_oidc_375ac979"
+       name           = "/projects/study/admins"
    }

  # vault_identity_group_alias.study_members will be created
+ resource "vault_identity_group_alias" "study_members" {
+       canonical_id   = (known after apply)
+       id             = (known after apply)
+       mount_accessor = "auth_oidc_375ac979"
+       name           = "/projects/study"
    }

  # vault_policy.study_dev will be created
+ resource "vault_policy" "study_dev" {
+       allow_overwrite = (known after apply)
+       id              = (known after apply)
+       name            = "study-dev"
+       policy          = <<-EOT
            path "secret/data/secretspec/study/dev/*" {
              capabilities = ["create", "read", "update", "delete", "list"]
            }
            
            path "secret/metadata/secretspec/study/dev/*" {
              capabilities = ["list", "read", "delete"]
            }
            
            path "secret/data/secretspec/study/preview/*" {
              capabilities = ["create", "read", "update", "delete", "list"]
            }
            
            path "secret/metadata/secretspec/study/preview/*" {
              capabilities = ["list", "read", "delete"]
            }
        EOT
    }

  # vault_policy.study_prod will be created
+ resource "vault_policy" "study_prod" {
+       allow_overwrite = (known after apply)
+       id              = (known after apply)
+       name            = "study-prod"
+       policy          = <<-EOT
            path "secret/data/secretspec/study/prod/*" {
              capabilities = ["create", "read", "update", "delete", "list"]
            }
            
            path "secret/metadata/secretspec/study/prod/*" {
              capabilities = ["list", "read", "delete"]
            }
            
            path "secret/data/secretspec/study/staging/*" {
              capabilities = ["create", "read", "update", "delete", "list"]
            }
            
            path "secret/metadata/secretspec/study/staging/*" {
              capabilities = ["list", "read", "delete"]
            }
            
            path "secret/data/secretspec/study/ci/*" {
              capabilities = ["create", "read", "update", "delete", "list"]
            }
            
            path "secret/metadata/secretspec/study/ci/*" {
              capabilities = ["list", "read", "delete"]
            }
        EOT
    }

Plan: 6 to add, 24 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p openbao
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p openbao
    

Plan: 6 to add, 24 to change, 0 to destroy.


5. project: google_play dir: tofu/google_play workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.33s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

6. project: sentry dir: tofu/sentry workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.40s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # sentry_key.cmu-study_study will be created
+ resource "sentry_key" "cmu-study_study" {
+       dsn                      = (sensitive value)
+       dsn_csp                  = (known after apply)
+       dsn_public               = (known after apply)
+       dsn_secret               = (sensitive value)
+       id                       = (known after apply)
+       javascript_loader_script = (known after apply)
+       name                     = "default"
+       organization             = "scottylabs"
+       project                  = "study"
+       project_id               = (known after apply)
+       public                   = (known after apply)
+       rate_limit_count         = (known after apply)
+       rate_limit_window        = (known after apply)
+       secret                   = (sensitive value)
    }

  # sentry_project.cmu-study_study will be created
+ resource "sentry_project" "cmu-study_study" {
+       client_security       = (known after apply)
+       digests_max_delay     = (known after apply)
+       digests_min_delay     = (known after apply)
+       features              = (known after apply)
+       filters               = (known after apply)
+       fingerprinting_rules  = (known after apply)
+       grouping_enhancements = (known after apply)
+       highlight_tags        = (known after apply)
+       id                    = (known after apply)
+       internal_id           = (known after apply)
+       name                  = "study"
+       organization          = "scottylabs"
+       resolve_age           = (known after apply)
+       slug                  = "study"
+       teams                 = [
+           "cmu-study",
        ]
    }

  # sentry_team.cmu-study will be created
+ resource "sentry_team" "cmu-study" {
+       has_access   = (known after apply)
+       id           = (known after apply)
+       internal_id  = (known after apply)
+       is_member    = (known after apply)
+       is_pending   = (known after apply)
+       name         = "CMU Study"
+       organization = "scottylabs"
+       slug         = "cmu-study"
+       team_id      = (known after apply)
    }

  # vault_kv_secret_v2.cmu-study_study_sentry_dsn will be created
+ resource "vault_kv_secret_v2" "cmu-study_study_sentry_dsn" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/SENTRY_DSN"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

Plan: 4 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p sentry
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p sentry
    

Plan: 4 to add, 0 to change, 0 to destroy.


7. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.34s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.ops_annadavi will be created
+ resource "google_cloud_identity_group_membership" "ops_annadavi" {
+       create_time = (known after apply)
+       group       = "groups/03fwokq036zsp1f"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "annadavi@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

  # google_cloud_identity_group_membership.tech_annadavi will be created
+ resource "google_cloud_identity_group_membership" "tech_annadavi" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "annadavi@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 2 to add, 0 to change, 0 to destroy.


8. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.41s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group.project_cmu-study_study will be created
+ resource "keycloak_group" "project_cmu-study_study" {
+       id        = (known after apply)
+       name      = "study"
+       parent_id = "9e6890c8-50a0-4b82-86b9-91791fffac25"
+       path      = (known after apply)
+       realm_id  = "scottylabs"
    }

  # keycloak_group.project_cmu-study_study_admins will be created
+ resource "keycloak_group" "project_cmu-study_study_admins" {
+       id        = (known after apply)
+       name      = "admins"
+       parent_id = (known after apply)
+       path      = (known after apply)
+       realm_id  = "scottylabs"
    }

  # keycloak_group_memberships.project_cmu-study_study will be created
+ resource "keycloak_group_memberships" "project_cmu-study_study" {
+       group_id = (known after apply)
+       id       = (known after apply)
+       members  = [
+           "annadavi",
        ]
+       realm_id = "scottylabs"
    }

  # keycloak_group_memberships.project_cmu-study_study_admins will be created
+ resource "keycloak_group_memberships" "project_cmu-study_study_admins" {
+       group_id = (known after apply)
+       id       = (known after apply)
+       members  = [
+           "annadavi",
        ]
+       realm_id = "scottylabs"
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "annadavi",
            # (35 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_openid_client.study will be created
+ resource "keycloak_openid_client" "study" {
+       access_token_lifespan                     = (known after apply)
+       access_type                               = "CONFIDENTIAL"
+       always_display_in_console                 = false
+       backchannel_logout_session_required       = true
+       client_authenticator_type                 = "client-secret"
+       client_id                                 = "study"
+       client_offline_session_idle_timeout       = (known after apply)
+       client_offline_session_max_lifespan       = (known after apply)
+       client_secret                             = (sensitive value)
+       client_secret_wo                          = (write-only attribute)
+       client_session_idle_timeout               = (known after apply)
+       client_session_max_lifespan               = (known after apply)
+       consent_required                          = (known after apply)
+       direct_access_grants_enabled              = false
+       display_on_consent_screen                 = (known after apply)
+       enabled                                   = true
+       exclude_issuer_from_auth_response         = (known after apply)
+       exclude_session_state_from_auth_response  = (known after apply)
+       frontchannel_logout_enabled               = (known after apply)
+       full_scope_allowed                        = true
+       id                                        = (known after apply)
+       implicit_flow_enabled                     = (known after apply)
+       import                                    = false
+       name                                      = "study"
+       oauth2_device_authorization_grant_enabled = false
+       oauth2_jwt_authorization_grant_enabled    = false
+       realm_id                                  = "scottylabs"
+       require_dpop_bound_tokens                 = (known after apply)
+       resource_server_id                        = (known after apply)
+       service_account_user_id                   = (known after apply)
+       service_accounts_enabled                  = (known after apply)
+       standard_flow_enabled                     = true
+       standard_token_exchange_enabled           = false
+       use_refresh_tokens                        = true
+       use_refresh_tokens_client_credentials     = false
+       valid_post_logout_redirect_uris           = (known after apply)
+       valid_redirect_uris                       = [
+           "https://oauth.scottylabs.org/oauth2/callback",
        ]
+       web_origins                               = [
+           "+",
        ]
    }

  # keycloak_openid_client.study_dev will be created
+ resource "keycloak_openid_client" "study_dev" {
+       access_token_lifespan                     = (known after apply)
+       access_type                               = "CONFIDENTIAL"
+       always_display_in_console                 = false
+       backchannel_logout_session_required       = true
+       client_authenticator_type                 = "client-secret"
+       client_id                                 = "study-dev"
+       client_offline_session_idle_timeout       = (known after apply)
+       client_offline_session_max_lifespan       = (known after apply)
+       client_secret                             = (sensitive value)
+       client_secret_wo                          = (write-only attribute)
+       client_session_idle_timeout               = (known after apply)
+       client_session_max_lifespan               = (known after apply)
+       consent_required                          = (known after apply)
+       direct_access_grants_enabled              = false
+       display_on_consent_screen                 = (known after apply)
+       enabled                                   = true
+       exclude_issuer_from_auth_response         = (known after apply)
+       exclude_session_state_from_auth_response  = (known after apply)
+       frontchannel_logout_enabled               = (known after apply)
+       full_scope_allowed                        = true
+       id                                        = (known after apply)
+       implicit_flow_enabled                     = (known after apply)
+       import                                    = false
+       name                                      = "study-dev"
+       oauth2_device_authorization_grant_enabled = false
+       oauth2_jwt_authorization_grant_enabled    = false
+       realm_id                                  = "scottylabs"
+       require_dpop_bound_tokens                 = (known after apply)
+       resource_server_id                        = (known after apply)
+       service_account_user_id                   = (known after apply)
+       service_accounts_enabled                  = (known after apply)
+       standard_flow_enabled                     = true
+       standard_token_exchange_enabled           = false
+       use_refresh_tokens                        = true
+       use_refresh_tokens_client_credentials     = false
+       valid_post_logout_redirect_uris           = (known after apply)
+       valid_redirect_uris                       = [
+           "http://localhost:8090/oauth2/callback",
        ]
+       web_origins                               = [
+           "+",
        ]
    }

  # keycloak_openid_client.study_staging will be created
+ resource "keycloak_openid_client" "study_staging" {
+       access_token_lifespan                     = (known after apply)
+       access_type                               = "CONFIDENTIAL"
+       always_display_in_console                 = false
+       backchannel_logout_session_required       = true
+       client_authenticator_type                 = "client-secret"
+       client_id                                 = "study-staging"
+       client_offline_session_idle_timeout       = (known after apply)
+       client_offline_session_max_lifespan       = (known after apply)
+       client_secret                             = (sensitive value)
+       client_secret_wo                          = (write-only attribute)
+       client_session_idle_timeout               = (known after apply)
+       client_session_max_lifespan               = (known after apply)
+       consent_required                          = (known after apply)
+       direct_access_grants_enabled              = false
+       display_on_consent_screen                 = (known after apply)
+       enabled                                   = true
+       exclude_issuer_from_auth_response         = (known after apply)
+       exclude_session_state_from_auth_response  = (known after apply)
+       frontchannel_logout_enabled               = (known after apply)
+       full_scope_allowed                        = true
+       id                                        = (known after apply)
+       implicit_flow_enabled                     = (known after apply)
+       import                                    = false
+       name                                      = "study-staging"
+       oauth2_device_authorization_grant_enabled = false
+       oauth2_jwt_authorization_grant_enabled    = false
+       realm_id                                  = "scottylabs"
+       require_dpop_bound_tokens                 = (known after apply)
+       resource_server_id                        = (known after apply)
+       service_account_user_id                   = (known after apply)
+       service_accounts_enabled                  = (known after apply)
+       standard_flow_enabled                     = true
+       standard_token_exchange_enabled           = false
+       use_refresh_tokens                        = true
+       use_refresh_tokens_client_credentials     = false
+       valid_post_logout_redirect_uris           = (known after apply)
+       valid_redirect_uris                       = [
+           "https://oauth.scottylabs.org/oauth2/callback",
        ]
+       web_origins                               = [
+           "+",
        ]
    }

  # keycloak_openid_group_membership_protocol_mapper.study will be created
+ resource "keycloak_openid_group_membership_protocol_mapper" "study" {
+       add_to_access_token        = true
+       add_to_id_token            = true
+       add_to_token_introspection = false
+       add_to_userinfo            = true
+       claim_name                 = "groups"
+       client_id                  = (known after apply)
+       full_path                  = true
+       id                         = (known after apply)
+       name                       = "groups"
+       realm_id                   = "scottylabs"
    }

  # keycloak_openid_group_membership_protocol_mapper.study_dev will be created
+ resource "keycloak_openid_group_membership_protocol_mapper" "study_dev" {
+       add_to_access_token        = true
+       add_to_id_token            = true
+       add_to_token_introspection = false
+       add_to_userinfo            = true
+       claim_name                 = "groups"
+       client_id                  = (known after apply)
+       full_path                  = true
+       id                         = (known after apply)
+       name                       = "groups"
+       realm_id                   = "scottylabs"
    }

  # keycloak_openid_group_membership_protocol_mapper.study_staging will be created
+ resource "keycloak_openid_group_membership_protocol_mapper" "study_staging" {
+       add_to_access_token        = true
+       add_to_id_token            = true
+       add_to_token_introspection = false
+       add_to_userinfo            = true
+       claim_name                 = "groups"
+       client_id                  = (known after apply)
+       full_path                  = true
+       id                         = (known after apply)
+       name                       = "groups"
+       realm_id                   = "scottylabs"
    }

  # vault_kv_secret_v2.study_dev_keycloak_realm will be created
+ resource "vault_kv_secret_v2" "study_dev_keycloak_realm" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/dev/KEYCLOAK_REALM"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_dev_keycloak_url will be created
+ resource "vault_kv_secret_v2" "study_dev_keycloak_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/dev/KEYCLOAK_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_dev_oauth_relay_url will be created
+ resource "vault_kv_secret_v2" "study_dev_oauth_relay_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/dev/OAUTH_RELAY_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_dev_oidc_client_id will be created
+ resource "vault_kv_secret_v2" "study_dev_oidc_client_id" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/dev/OIDC_CLIENT_ID"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_dev_oidc_client_secret will be created
+ resource "vault_kv_secret_v2" "study_dev_oidc_client_secret" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/dev/OIDC_CLIENT_SECRET"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_dev_project_admin_group will be created
+ resource "vault_kv_secret_v2" "study_dev_project_admin_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/dev/PROJECT_ADMIN_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_dev_project_group will be created
+ resource "vault_kv_secret_v2" "study_dev_project_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/dev/PROJECT_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_preview_keycloak_realm will be created
+ resource "vault_kv_secret_v2" "study_preview_keycloak_realm" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/preview/KEYCLOAK_REALM"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_preview_keycloak_url will be created
+ resource "vault_kv_secret_v2" "study_preview_keycloak_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/preview/KEYCLOAK_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_preview_oauth_relay_url will be created
+ resource "vault_kv_secret_v2" "study_preview_oauth_relay_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/preview/OAUTH_RELAY_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_preview_oidc_client_id will be created
+ resource "vault_kv_secret_v2" "study_preview_oidc_client_id" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/preview/OIDC_CLIENT_ID"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_preview_oidc_client_secret will be created
+ resource "vault_kv_secret_v2" "study_preview_oidc_client_secret" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/preview/OIDC_CLIENT_SECRET"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_preview_project_admin_group will be created
+ resource "vault_kv_secret_v2" "study_preview_project_admin_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/preview/PROJECT_ADMIN_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_preview_project_group will be created
+ resource "vault_kv_secret_v2" "study_preview_project_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/preview/PROJECT_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_prod_keycloak_realm will be created
+ resource "vault_kv_secret_v2" "study_prod_keycloak_realm" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/KEYCLOAK_REALM"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_prod_keycloak_url will be created
+ resource "vault_kv_secret_v2" "study_prod_keycloak_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/KEYCLOAK_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_prod_oauth_relay_url will be created
+ resource "vault_kv_secret_v2" "study_prod_oauth_relay_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/OAUTH_RELAY_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_prod_oidc_client_id will be created
+ resource "vault_kv_secret_v2" "study_prod_oidc_client_id" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/OIDC_CLIENT_ID"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_prod_oidc_client_secret will be created
+ resource "vault_kv_secret_v2" "study_prod_oidc_client_secret" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/OIDC_CLIENT_SECRET"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_prod_project_admin_group will be created
+ resource "vault_kv_secret_v2" "study_prod_project_admin_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/PROJECT_ADMIN_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_prod_project_group will be created
+ resource "vault_kv_secret_v2" "study_prod_project_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/prod/PROJECT_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_staging_keycloak_realm will be created
+ resource "vault_kv_secret_v2" "study_staging_keycloak_realm" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/staging/KEYCLOAK_REALM"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_staging_keycloak_url will be created
+ resource "vault_kv_secret_v2" "study_staging_keycloak_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/staging/KEYCLOAK_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_staging_oauth_relay_url will be created
+ resource "vault_kv_secret_v2" "study_staging_oauth_relay_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/staging/OAUTH_RELAY_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_staging_oidc_client_id will be created
+ resource "vault_kv_secret_v2" "study_staging_oidc_client_id" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/staging/OIDC_CLIENT_ID"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_staging_oidc_client_secret will be created
+ resource "vault_kv_secret_v2" "study_staging_oidc_client_secret" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/staging/OIDC_CLIENT_SECRET"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_staging_project_admin_group will be created
+ resource "vault_kv_secret_v2" "study_staging_project_admin_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/staging/PROJECT_ADMIN_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.study_staging_project_group will be created
+ resource "vault_kv_secret_v2" "study_staging_project_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/study/staging/PROJECT_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

Plan: 38 to add, 1 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 38 to add, 1 to change, 0 to destroy.


9. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.28s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_channel_permission.cmu-study_ch0 will be created
+ resource "discord_channel_permission" "cmu-study_ch0" {
+       allow        = 1024
+       channel_id   = "1512270354943578132"
+       id           = (known after apply)
+       overwrite_id = (known after apply)
+       type         = "role"
    }

  # discord_member_roles.annadavi will be created
+ resource "discord_member_roles" "annadavi" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "991766392590311455"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1519792320755662928"
        }
+       role {
+           has_role = true
+           role_id  = (known after apply)
        }
    }

  # discord_role.cmu-study will be created
+ resource "discord_role" "cmu-study" {
+       hoist       = false
+       id          = (known after apply)
+       managed     = (known after apply)
+       mentionable = false
+       name        = "CMU Study"
+       permissions = 0
+       position    = (known after apply)
+       server_id   = "1353118209024331948"
    }

Plan: 3 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 3 to add, 0 to change, 0 to destroy.


10. project: git dir: tofu/git workspace: default

Plan Error

Show Output
running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu plan -input=false -refresh -out /var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git/git-default.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git': exit status 1
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
random_password.kennel_webhook_secret: Refreshing state... [id=none]
github_repository.quest_quest: Refreshing state... [id=quest]
github_team.cmu-housing: Refreshing state... [id=17937919]
github_repository.slai_cmugpt_sms_surface: Refreshing state... [id=cmugpt-sms-surface]
github_repository.cmu-insta_insta: Refreshing state... [id=insta]
github_team.cmu-insta: Refreshing state... [id=19240311]
github_repository.cmu-maps_maps: Refreshing state... [id=maps]
github_repository.uia_scottylabs_org: Refreshing state... [id=scottylabs-org]
github_repository.slai_cmugpt_agent: Refreshing state... [id=cmugpt-agent]
github_repository.cbp_dalmatian: Refreshing state... [id=dalmatian]
github_repository.lost-and-found_lost_and_found_v2: Refreshing state... [id=lost-and-found-v2]
github_team.terrier: Refreshing state... [id=12028577]
github_repository.cmu-housing_housing: Refreshing state... [id=housing]
github_repository.cmu-cal_cal: Refreshing state... [id=cal]
github_team.occ: Refreshing state... [id=19240310]
github_team.devops: Refreshing state... [id=14706218]
github_team.lost-and-found: Refreshing state... [id=4088608]
github_team.cmu-maps: Refreshing state... [id=10872360]
github_repository.devops_infrastructure: Refreshing state... [id=infrastructure]
github_repository.occ_occ: Refreshing state... [id=occ]
github_repository.devops_documentation: Refreshing state... [id=documentation]
github_repository.cmu-courses_courses: Refreshing state... [id=courses]
github_repository.slai_cmugpt_surface: Refreshing state... [id=cmugpt-surface]
github_repository.cmu-courses_internet_archive: Refreshing state... [id=internet-archive]
github_repository.cmu-eats_dining_api: Refreshing state... [id=dining-api]
github_repository.terrier_terrier: Refreshing state... [id=terrier]
github_team.cmu-courses: Refreshing state... [id=3643907]
github_team.cmu-eats: Refreshing state... [id=16807600]
github_repository.devops_link_shortener: Refreshing state... [id=link-shortener]
github_repository.cbp_groupme_mirror: Refreshing state... [id=groupme-mirror]
github_repository.uia_components: Refreshing state... [id=components]
github_team.quest: Refreshing state... [id=17330197]
github_repository.devops_kennel: Refreshing state... [id=kennel]
github_repository.devops_collie: Refreshing state... [id=collie]
github_repository.cbp_bus_sign: Refreshing state... [id=bus-sign]
github_repository.cbp_discord_verify: Refreshing state... [id=discord-verify]
github_team.cbp: Refreshing state... [id=17170099]
github_team.tartan-vote: Refreshing state... [id=17701833]
github_team.cmu-cal: Refreshing state... [id=15669112]
github_team.uia: Refreshing state... [id=17529763]
github_repository.devops_governance: Refreshing state... [id=governance]
github_team.slai: Refreshing state... [id=17571328]
github_repository.cmu-eats_eats: Refreshing state... [id=eats]
github_repository.tartan-vote_tartan_vote: Refreshing state... [id=tartan-vote]
github_repository.slai_mcp_server: Refreshing state... [id=mcp-server]
forgejo_repository.devops_infrastructure: Refreshing state... [name=infrastructure]
forgejo_repository.devops_collie: Refreshing state... [name=collie]
forgejo_repository.quest_quest: Refreshing state... [name=quest]
forgejo_repository.cmu-cal_cal: Refreshing state... [name=cal]
forgejo_repository.tartan-vote_tartan_vote: Refreshing state... [name=tartan-vote]
forgejo_repository.uia_scottylabs_org: Refreshing state... [name=scottylabs-org]
forgejo_repository.cmu-courses_internet_archive: Refreshing state... [name=internet-archive]
forgejo_repository.devops_kennel: Refreshing state... [name=kennel]
forgejo_repository.devops_link_shortener: Refreshing state... [name=link-shortener]
forgejo_repository.slai_mcp_server: Refreshing state... [name=mcp-server]
forgejo_repository.cbp_dalmatian: Refreshing state... [name=dalmatian]
forgejo_repository.lost-and-found_lost_and_found_v2: Refreshing state... [name=lost-and-found-v2]
forgejo_repository.cmu-courses_courses: Refreshing state... [name=courses]
forgejo_repository.cbp_groupme_mirror: Refreshing state... [name=groupme-mirror]
forgejo_repository.cmu-eats_eats: Refreshing state... [name=eats]
forgejo_repository.cbp_bus_sign: Refreshing state... [name=bus-sign]
forgejo_repository.occ_occ: Refreshing state... [name=occ]
data.forgejo_organization.this: Reading...
data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs]
forgejo_repository.slai_cmugpt_sms_surface: Refreshing state... [name=cmugpt-sms-surface]
forgejo_repository.cmu-eats_dining_api: Refreshing state... [name=dining-api]
forgejo_repository.devops_documentation: Refreshing state... [name=documentation]
forgejo_repository.cmu-insta_insta: Refreshing state... [name=insta]
forgejo_repository.slai_cmugpt_surface: Refreshing state... [name=cmugpt-surface]
forgejo_repository.terrier_terrier: Refreshing state... [name=terrier]
forgejo_repository.slai_cmugpt_agent: Refreshing state... [name=cmugpt-agent]
forgejo_repository.cbp_discord_verify: Refreshing state... [name=discord-verify]
forgejo_repository.devops_governance: Refreshing state... [name=governance]
forgejo_repository.cmu-housing_housing: Refreshing state... [name=housing]
forgejo_repository.uia_components: Refreshing state... [name=components]
forgejo_repository.cmu-maps_maps: Refreshing state... [name=maps]
data.external.identity_acchs: Reading...
data.external.identity_nikhilr: Reading...
data.external.identity_dcrodrig: Reading...
data.external.identity_xboxbedrock: Reading...
data.external.identity_acchs: Read complete after 0s [id=-]
data.external.identity_brightz: Reading...
data.external.identity_dcrodrig: Read complete after 0s [id=-]
data.external.identity_qianxuege: Reading...
data.external.identity_nikhilr: Read complete after 0s [id=-]
data.external.identity_maybe_yiyi: Reading...
data.external.identity_dkou: Reading...
data.external.identity_xboxbedrock: Read complete after 1s [id=-]
data.external.identity_jc_965: Reading...
data.external.identity_qianxuege: Read complete after 1s [id=-]
data.external.identity_gostmeaper: Reading...
data.external.identity_maybe_yiyi: Read complete after 1s [id=-]
data.external.identity_kothurvidur: Reading...
data.external.identity_brightz: Read complete after 1s [id=-]
data.external.identity_manrajs: Reading...
data.external.identity_emmawong: Reading...
data.external.identity_dkou: Read complete after 1s [id=-]
data.external.identity_jalenluorion: Reading...
data.external.identity_jc_965: Read complete after 0s [id=-]
data.external.identity_thesuperrl: Reading...
data.external.identity_kothurvidur: Read complete after 0s [id=-]
data.external.identity_riztofu: Reading...
data.external.identity_manrajs: Read complete after 0s [id=-]
data.external.identity_arom1a: Reading...
data.external.identity_gostmeaper: Read complete after 0s [id=-]
data.external.identity_liamwalk: Reading...
data.external.identity_emmawong: Read complete after 0s [id=-]
data.external.identity_stringatlas: Reading...
data.external.identity_jalenluorion: Read complete after 0s [id=-]
data.external.identity_rzhu3: Reading...
data.external.identity_thesuperrl: Read complete after 0s [id=-]
data.external.identity_amzoeee: Reading...
data.external.identity_arom1a: Read complete after 0s [id=-]
data.external.identity_krishsax: Reading...
data.external.identity_liamwalk: Read complete after 0s [id=-]
data.external.identity_jw8: Reading...
data.external.identity_riztofu: Read complete after 0s [id=-]
data.external.identity_rhsen: Reading...
data.external.identity_stringatlas: Read complete after 0s [id=-]
data.external.identity_jessica_lr_song: Reading...
data.external.identity_rzhu3: Read complete after 0s [id=-]
data.external.identity_anish: Reading...
data.external.identity_amzoeee: Read complete after 0s [id=-]
data.external.identity_annadavi: Reading...
data.external.identity_jw8: Read complete after 0s [id=-]
data.external.identity_siddharthr: Reading...
data.external.identity_krishsax: Read complete after 0s [id=-]
data.external.identity_gabrielx: Reading...
data.external.identity_rhsen: Read complete after 0s [id=-]
data.external.identity_kluoxx: Reading...
data.external.identity_jessica_lr_song: Read complete after 0s [id=-]
data.external.identity_tentype: Reading...
data.external.identity_anish: Read complete after 0s [id=-]
data.external.identity_kritdass: Reading...
data.external.identity_annadavi: Read complete after 0s [id=-]
data.external.identity_jiyas: Reading...
data.external.identity_kluoxx: Read complete after 1s [id=-]
data.external.identity_zoexue: Reading...
data.external.identity_siddharthr: Read complete after 1s [id=-]
data.external.identity_kechezon: Reading...
data.external.identity_gabrielx: Read complete after 1s [id=-]
data.external.identity_etashjha: Reading...
data.external.identity_tentype: Read complete after 1s [id=-]
github_actions_repository_permissions.slai_cmugpt_sms_surface_actions: Refreshing state... [id=cmugpt-sms-surface]
data.external.identity_kritdass: Read complete after 1s [id=-]
github_actions_repository_permissions.slai_cmugpt_agent_actions: Refreshing state... [id=cmugpt-agent]
data.external.identity_jiyas: Read complete after 1s [id=-]
github_actions_repository_permissions.lost-and-found_lost_and_found_v2_actions: Refreshing state... [id=lost-and-found-v2]
github_actions_repository_permissions.quest_quest_actions: Refreshing state... [id=quest]
data.external.identity_zoexue: Read complete after 0s [id=-]
github_actions_repository_permissions.cmu-maps_maps_actions: Refreshing state... [id=maps]
data.external.identity_kechezon: Read complete after 0s [id=-]
github_actions_repository_permissions.cmu-insta_insta_actions: Refreshing state... [id=insta]
data.external.identity_etashjha: Read complete after 0s [id=-]
github_actions_repository_permissions.uia_scottylabs_org_actions: Refreshing state... [id=scottylabs-org]
github_actions_repository_permissions.cbp_dalmatian_actions: Refreshing state... [id=dalmatian]
github_actions_repository_permissions.cmu-housing_housing_actions: Refreshing state... [id=housing]
github_actions_repository_permissions.cmu-cal_cal_actions: Refreshing state... [id=cal]
github_actions_repository_permissions.cmu-courses_courses_actions: Refreshing state... [id=courses]
github_actions_repository_permissions.slai_cmugpt_surface_actions: Refreshing state... [id=cmugpt-surface]
github_actions_repository_permissions.cmu-courses_internet_archive_actions: Refreshing state... [id=internet-archive]
github_actions_repository_permissions.devops_infrastructure_actions: Refreshing state... [id=infrastructure]
github_actions_repository_permissions.occ_occ_actions: Refreshing state... [id=occ]
github_actions_repository_permissions.devops_documentation_actions: Refreshing state... [id=documentation]
github_actions_repository_permissions.cbp_groupme_mirror_actions: Refreshing state... [id=groupme-mirror]
github_actions_repository_permissions.uia_components_actions: Refreshing state... [id=components]
github_actions_repository_permissions.cmu-eats_dining_api_actions: Refreshing state... [id=dining-api]
github_actions_repository_permissions.terrier_terrier_actions: Refreshing state... [id=terrier]
github_actions_repository_permissions.devops_collie_actions: Refreshing state... [id=collie]
github_actions_repository_permissions.cbp_discord_verify_actions: Refreshing state... [id=discord-verify]
github_actions_repository_permissions.devops_link_shortener_actions: Refreshing state... [id=link-shortener]
restapi_object.devops_infrastructure_push_mirror: Refreshing state... [id=remote_mirror_oYtFpKVzTaZ]
restapi_object.devops_infrastructure_kennel_webhook: Refreshing state... [id=26]
restapi_object.devops_collie_push_mirror: Refreshing state... [id=remote_mirror_ljvVejNFXaS]
restapi_object.devops_collie_kennel_webhook: Refreshing state... [id=17]
restapi_object.quest_quest_kennel_webhook: Refreshing state... [id=10]
restapi_object.quest_quest_push_mirror: Refreshing state... [id=remote_mirror_U9d9BPAon29]
restapi_object.cmu-cal_cal_push_mirror: Refreshing state... [id=remote_mirror_CO243uIONoq]
restapi_object.cmu-cal_cal_kennel_webhook: Refreshing state... [id=20]
restapi_object.tartan-vote_tartan_vote_kennel_webhook: Refreshing state... [id=24]
restapi_object.tartan-vote_tartan_vote_push_mirror: Refreshing state... [id=remote_mirror_EUFq4sFSnH5]
restapi_object.tartan-vote_tartan_vote_docs_webhook: Refreshing state... [id=25]
restapi_object.uia_scottylabs_org_kennel_webhook: Refreshing state... [id=29]
restapi_object.uia_scottylabs_org_push_mirror: Refreshing state... [id=remote_mirror_h6yN6QdS1lD]
restapi_object.cmu-courses_internet_archive_push_mirror: Refreshing state... [id=remote_mirror_58X__Yfs0bR]
restapi_object.devops_link_shortener_kennel_webhook: Refreshing state... [id=8]
restapi_object.devops_link_shortener_push_mirror: Refreshing state... [id=remote_mirror_cp8daMpt82v]
restapi_object.devops_kennel_kennel_webhook: Refreshing state... [id=14]
restapi_object.devops_kennel_push_mirror: Refreshing state... [id=remote_mirror_cT9psJkhTQ4]
restapi_object.slai_mcp_server_push_mirror: Refreshing state... [id=remote_mirror_Tqh2y0R_vDp]
restapi_object.slai_mcp_server_kennel_webhook: Refreshing state... [id=13]
restapi_object.cbp_dalmatian_kennel_webhook: Refreshing state... [id=6]
restapi_object.cbp_dalmatian_push_mirror: Refreshing state... [id=remote_mirror_qzlCpLK153F]
restapi_object.cbp_dalmatian_docs_webhook: Refreshing state... [id=7]
restapi_object.lost-and-found_lost_and_found_v2_push_mirror: Refreshing state... [id=remote_mirror_n_UZBFU-CY4]
restapi_object.cmu-courses_courses_push_mirror: Refreshing state... [id=remote_mirror_WMR9yQwKFDF]
restapi_object.cmu-courses_courses_kennel_webhook: Refreshing state... [id=11]
restapi_object.cbp_groupme_mirror_push_mirror: Refreshing state... [id=remote_mirror_U61HJHB9VGE]
restapi_object.cmu-eats_eats_push_mirror: Refreshing state... [id=remote_mirror_sjYwMA6NPgh]
restapi_object.cmu-eats_eats_kennel_webhook: Refreshing state... [id=31]
restapi_object.cbp_bus_sign_push_mirror: Refreshing state... [id=remote_mirror_MkKxQ2vA9pc]
restapi_object.cbp_bus_sign_kennel_webhook: Refreshing state... [id=21]
forgejo_team.devops: Refreshing state... [name=devops]
forgejo_team.bus-sign_leads: Refreshing state... [name=bus-sign-leads]
forgejo_team.courses_leads: Refreshing state... [name=courses-leads]
forgejo_team.cmu-courses: Refreshing state... [name=cmu-courses]
forgejo_team.scottylabs-org_leads: Refreshing state... [name=scottylabs-org-leads]
forgejo_team.dalmatian_leads: Refreshing state... [name=dalmatian-leads]
forgejo_team.cmu-cal: Refreshing state... [name=cmu-cal]
forgejo_team.cmugpt-surface_leads: Refreshing state... [name=cmugpt-surface-leads]
forgejo_team.tartan-vote_leads: Refreshing state... [name=tartan-vote-leads]
forgejo_team.insta_leads: Refreshing state... [name=insta-leads]
forgejo_team.link-shortener_leads: Refreshing state... [name=link-shortener-leads]
forgejo_team.lost-and-found-v2_leads: Refreshing state... [name=lost-and-found-v2-leads]
forgejo_team.discord-verify_leads: Refreshing state... [name=discord-verify-leads]
forgejo_team.cmugpt-sms-surface_leads: Refreshing state... [name=cmugpt-sms-surface-leads]
forgejo_team.quest: Refreshing state... [name=quest]
forgejo_team.lost-and-found: Refreshing state... [name=lost-and-found]
forgejo_team.tartan-vote: Refreshing state... [name=tartan-vote]
forgejo_team.occ-main_leads: Refreshing state... [name=occ-main-leads]
forgejo_team.documentation_leads: Refreshing state... [name=documentation-leads]
forgejo_team.cal_leads: Refreshing state... [name=cal-leads]
forgejo_team.cmu-housing: Refreshing state... [name=cmu-housing]
forgejo_team.terrier: Refreshing state... [name=terrier]
forgejo_team.cmugpt-mcp-server_leads: Refreshing state... [name=cmugpt-mcp-server-leads]
forgejo_team.cmugpt-agent_leads: Refreshing state... [name=cmugpt-agent-leads]
forgejo_team.groupme-mirror_leads: Refreshing state... [name=groupme-mirror-leads]
forgejo_team.cmu-insta: Refreshing state... [name=cmu-insta]
forgejo_team.cmu-courses_leads: Refreshing state... [name=cmu-courses-leads]
forgejo_team.components_leads: Refreshing state... [name=components-leads]
forgejo_team.cbp: Refreshing state... [name=cbp]
forgejo_team.housing_leads: Refreshing state... [name=housing-leads]
forgejo_team.devops_leads: Refreshing state... [name=devops-leads]
forgejo_team.uia: Refreshing state... [name=uia]
forgejo_team.terrier_leads: Refreshing state... [name=terrier-leads]
forgejo_team.slai: Refreshing state... [name=slai]
forgejo_team.maps_leads: Refreshing state... [name=maps-leads]
forgejo_team.quest_leads: Refreshing state... [name=quest-leads]
forgejo_team.cmu-maps: Refreshing state... [name=cmu-maps]
forgejo_team.eats_leads: Refreshing state... [name=eats-leads]
forgejo_team.occ: Refreshing state... [name=occ]
forgejo_team.cmu-eats: Refreshing state... [name=cmu-eats]
restapi_object.occ_occ_push_mirror: Refreshing state... [id=remote_mirror_moBLq1YG1qz]
restapi_object.devops_documentation_push_mirror: Refreshing state... [id=remote_mirror_izmsqaNB6Wx]
restapi_object.slai_cmugpt_sms_surface_push_mirror: Refreshing state... [id=remote_mirror__fkBkASN2jA]
restapi_object.slai_cmugpt_sms_surface_kennel_webhook: Refreshing state... [id=19]
restapi_object.cmu-eats_dining_api_push_mirror: Refreshing state... [id=remote_mirror_a4MsQ5rO1_Z]
restapi_object.slai_cmugpt_surface_kennel_webhook: Refreshing state... [id=23]
restapi_object.slai_cmugpt_surface_push_mirror: Refreshing state... [id=remote_mirror_cPTQN6S5SpB]
restapi_object.cmu-insta_insta_push_mirror: Refreshing state... [id=remote_mirror_891nu6XQ1us]
restapi_object.terrier_terrier_push_mirror: Refreshing state... [id=remote_mirror_YopDgZksZv5]
restapi_object.terrier_terrier_kennel_webhook: Refreshing state... [id=12]
restapi_object.slai_cmugpt_agent_push_mirror: Refreshing state... [id=remote_mirror_kNY6rVcxdyr]
restapi_object.slai_cmugpt_agent_kennel_webhook: Refreshing state... [id=28]
restapi_object.devops_governance_kennel_webhook: Refreshing state... [id=22]
restapi_object.devops_governance_push_mirror: Refreshing state... [id=remote_mirror_lt9JzkX1jMq]
restapi_object.cbp_discord_verify_push_mirror: Refreshing state... [id=remote_mirror_3MV7Up6cY1X]
restapi_object.cbp_discord_verify_kennel_webhook: Refreshing state... [id=18]
restapi_object.cmu-housing_housing_kennel_webhook: Refreshing state... [id=16]
restapi_object.cmu-housing_housing_docs_webhook: Refreshing state... [id=15]
restapi_object.cmu-housing_housing_push_mirror: Refreshing state... [id=remote_mirror_uoV2HG4aiOT]
restapi_object.uia_components_push_mirror: Refreshing state... [id=remote_mirror_Kwwa8TDnTgc]
restapi_object.uia_components_kennel_webhook: Refreshing state... [id=27]
restapi_object.cmu-maps_maps_kennel_webhook: Refreshing state... [id=9]
restapi_object.cmu-maps_maps_push_mirror: Refreshing state... [id=remote_mirror_e4_FUnvoslJ]
github_actions_repository_permissions.devops_kennel_actions: Refreshing state... [id=kennel]
github_team_membership.occ_acchs: Refreshing state... [id=19240310:acchs-at-cmu]
github_team_membership.occ_dcrodrig: Refreshing state... [id=19240310:newhacker1746]
github_team_membership.cmu-housing_nikhilr: Refreshing state... [id=17937919:ecstaticpilot]
github_team_membership.cbp_nikhilr: Refreshing state... [id=17170099:ecstaticpilot]
github_team_membership.quest_xboxbedrock: Refreshing state... [id=17330197:xboxbedrock]
github_team_membership.cbp_xboxbedrock: Refreshing state... [id=17170099:xboxbedrock]
github_team_membership.tartan-vote_xboxbedrock: Refreshing state... [id=17701833:xboxbedrock]
github_team_membership.cmu-housing_xboxbedrock: Refreshing state... [id=17937919:xboxbedrock]
github_team_membership.devops_xboxbedrock: Refreshing state... [id=14706218:xboxbedrock]
github_team_membership.occ_xboxbedrock: Refreshing state... [id=19240310:xboxbedrock]
github_team_membership.slai_xboxbedrock: Refreshing state... [id=17571328:xboxbedrock]
github_team_membership.slai_qianxuege: Refreshing state... [id=17571328:qianxuege]
github_team_membership.cbp_maybe_yiyi: Refreshing state... [id=17170099:maybe-yiyi]
github_team_membership.tartan-vote_maybe_yiyi: Refreshing state... [id=17701833:maybe-yiyi]
github_team_membership.quest_brightz: Refreshing state... [id=17330197:bright-zheng]
github_actions_repository_permissions.cbp_bus_sign_actions: Refreshing state... [id=bus-sign]
github_team_membership.cmu-courses_dkou: Refreshing state... [id=3643907:sharkree]
github_team_membership.terrier_dkou: Refreshing state... [id=12028577:sharkree]
github_team_membership.slai_jc_965: Refreshing state... [id=17571328:jc-965]
github_team_membership.terrier_kothurvidur: Refreshing state... [id=12028577:kothurvidur]
github_team_membership.cmu-housing_manrajs: Refreshing state... [id=17937919:manraj10]
github_team_membership.cmu-housing_gostmeaper: Refreshing state... [id=17937919:gostmeaper]
github_team_membership.cmu-housing_emmawong: Refreshing state... [id=17937919:emw868]
github_team_membership.cmu-cal_jalenluorion: Refreshing state... [id=15669112:jalenluorion]
github_team_membership.devops_thesuperrl: Refreshing state... [id=14706218:thesuperrl]
github_team_membership.cmu-eats_arom1a: Refreshing state... [id=16807600:arom1a]
github_team_membership.cbp_liamwalk: Refreshing state... [id=17170099:pandaaured]
github_team_membership.tartan-vote_riztofu: Refreshing state... [id=17701833:riztofu]
github_team_membership.cmu-housing_stringatlas: Refreshing state... [id=17937919:stringatlas]
github_team_membership.slai_rzhu3: Refreshing state... [id=17571328:richard-31415]
github_team_membership.terrier_amzoeee: Refreshing state... [id=12028577:amzoeee]
github_team_membership.cmu-housing_amzoeee: Refreshing state... [id=17937919:amzoeee]
github_team_membership.quest_jw8: Refreshing state... [id=17330197:jeffreyyijiwang]
github_team_membership.slai_krishsax: Refreshing state... [id=17571328:krishsaxena]
github_team_membership.cmu-housing_rhsen: Refreshing state... [id=17937919:rohansen08]
github_team_membership.cmu-housing_jessica_lr_song: Refreshing state... [id=17937919:jessica-lr-song]
github_team_membership.slai_anish: Refreshing state... [id=17571328:ap-1]
github_team_membership.terrier_anish: Refreshing state... [id=12028577:ap-1]
github_team_membership.cmu-housing_anish: Refreshing state... [id=17937919:ap-1]
github_team_membership.cmu-cal_anish: Refreshing state... [id=15669112:ap-1]
github_team_membership.occ_anish: Refreshing state... [id=19240310:ap-1]
github_team_membership.tartan-vote_anish: Refreshing state... [id=17701833:ap-1]
github_team_membership.quest_anish: Refreshing state... [id=17330197:ap-1]
github_team_membership.cmu-courses_anish: Refreshing state... [id=3643907:ap-1]
github_team_membership.cmu-maps_anish: Refreshing state... [id=10872360:ap-1]
github_team_membership.lost-and-found_anish: Refreshing state... [id=4088608:ap-1]
github_team_membership.devops_anish: Refreshing state... [id=14706218:ap-1]
github_team_membership.cbp_anish: Refreshing state... [id=17170099:ap-1]
github_team_membership.uia_anish: Refreshing state... [id=17529763:ap-1]
github_team_membership.cmu-housing_kluoxx: Refreshing state... [id=17937919:luokaden888]
github_team_membership.cmu-maps_siddharthr: Refreshing state... [id=10872360:overthrowing]
github_team_membership.cmu-courses_gabrielx: Refreshing state... [id=3643907:gabriel-xu]
github_team_membership.cmu-housing_tentype: Refreshing state... [id=17937919:tentype]
github_team_membership.cbp_tentype: Refreshing state... [id=17170099:tentype]
github_team_membership.tartan-vote_tentype: Refreshing state... [id=17701833:tentype]
github_team_membership.terrier_kritdass: Refreshing state... [id=12028577:kritdass]
github_team_membership.slai_jiyas: Refreshing state... [id=17571328:jiya2301]
github_actions_repository_permissions.devops_governance_actions: Refreshing state... [id=governance]
github_team_membership.cmu-housing_zoexue: Refreshing state... [id=17937919:o-zox-o]
github_team_membership.quest_kechezon: Refreshing state... [id=17330197:kechezon]
github_team_membership.terrier_etashjha: Refreshing state... [id=12028577:etashj]
github_team_membership.cmu-courses_etashjha: Refreshing state... [id=3643907:etashj]
github_team_membership.cmu-insta_etashjha: Refreshing state... [id=19240311:etashj]
github_actions_repository_permissions.cmu-eats_eats_actions: Refreshing state... [id=eats]
github_actions_repository_permissions.tartan-vote_tartan_vote_actions: Refreshing state... [id=tartan-vote]
github_actions_repository_permissions.slai_mcp_server_actions: Refreshing state... [id=mcp-server]
github_repository_deploy_key.devops_infrastructure_mirror_deploy_key: Refreshing state... [id=infrastructure:159665343]
github_repository_deploy_key.devops_collie_mirror_deploy_key: Refreshing state... [id=collie:159665332]
github_repository_deploy_key.quest_quest_mirror_deploy_key: Refreshing state... [id=quest:159665323]
github_repository_deploy_key.cmu-cal_cal_mirror_deploy_key: Refreshing state... [id=cal:159665344]
github_repository_deploy_key.tartan-vote_tartan_vote_mirror_deploy_key: Refreshing state... [id=tartan-vote:159665350]
github_repository_deploy_key.uia_scottylabs_org_mirror_deploy_key: Refreshing state... [id=scottylabs-org:159665361]
github_repository_deploy_key.cmu-courses_internet_archive_mirror_deploy_key: Refreshing state... [id=internet-archive:159665356]
github_repository_deploy_key.devops_link_shortener_mirror_deploy_key: Refreshing state... [id=link-shortener:159665345]
github_repository_deploy_key.devops_kennel_mirror_deploy_key: Refreshing state... [id=kennel:159665328]
github_repository_deploy_key.slai_mcp_server_mirror_deploy_key: Refreshing state... [id=mcp-server:159665358]
github_repository_deploy_key.cbp_dalmatian_mirror_deploy_key: Refreshing state... [id=dalmatian:159665367]
github_repository_deploy_key.lost-and-found_lost_and_found_v2_mirror_deploy_key: Refreshing state... [id=lost-and-found-v2:159665337]
github_repository_deploy_key.cmu-courses_courses_mirror_deploy_key: Refreshing state... [id=courses:159665327]
github_repository_deploy_key.cbp_groupme_mirror_mirror_deploy_key: Refreshing state... [id=groupme-mirror:159665366]
github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Refreshing state... [id=eats:161775995]
github_repository_deploy_key.cbp_bus_sign_mirror_deploy_key: Refreshing state... [id=bus-sign:159665342]
forgejo_team_repository.devops_link_shortener: Refreshing state...
forgejo_team_member.devops_thesuperrl: Refreshing state...
forgejo_team_repository.devops_kennel: Refreshing state...
forgejo_team_member.devops_anish: Refreshing state...
forgejo_team_member.devops_xboxbedrock: Refreshing state...
forgejo_team_repository.devops_governance: Refreshing state...
forgejo_team_repository.devops_infrastructure: Refreshing state...
forgejo_team_repository.devops_collie: Refreshing state...
forgejo_team_repository.devops_documentation: Refreshing state...
forgejo_team_member.bus-sign_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.cmu-courses_dkou: Refreshing state...
forgejo_team_member.cmu-courses_gabrielx: Refreshing state...
forgejo_team_member.bus-sign_leads_anish: Refreshing state...
forgejo_team_member.bus-sign_leads_tentype: Refreshing state...
forgejo_team_member.cmu-courses_etashjha: Refreshing state...
forgejo_team_member.cmu-courses_anish: Refreshing state...
forgejo_team_repository.cmu-courses_internet_archive: Refreshing state...
forgejo_team_repository.cmu-courses_courses: Refreshing state...
forgejo_team_member.courses_leads_anish: Refreshing state...
forgejo_team_member.courses_leads_etashjha: Refreshing state...
forgejo_team_member.scottylabs-org_leads_anish: Refreshing state...
forgejo_team_member.dalmatian_leads_tentype: Refreshing state...
forgejo_team_member.dalmatian_leads_anish: Refreshing state...
forgejo_team_member.dalmatian_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.cmu-cal_anish: Refreshing state...
forgejo_team_repository.cmu-cal_cal: Refreshing state...
forgejo_team_member.cmu-cal_jalenluorion: Refreshing state...
forgejo_team_member.cmugpt-surface_leads_anish: Refreshing state...
forgejo_team_member.cmugpt-surface_leads_krishsax: Refreshing state...
forgejo_team_member.tartan-vote_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.tartan-vote_leads_anish: Refreshing state...
forgejo_team_member.insta_leads_etashjha: Refreshing state...
forgejo_team_member.lost-and-found-v2_leads_anish: Refreshing state...
forgejo_team_member.link-shortener_leads_anish: Refreshing state...
forgejo_team_member.link-shortener_leads_thesuperrl: Refreshing state...
forgejo_team_member.cmugpt-sms-surface_leads_krishsax: Refreshing state...
forgejo_team_member.cmugpt-sms-surface_leads_anish: Refreshing state...
forgejo_team_member.discord-verify_leads_tentype: Refreshing state...
forgejo_team_member.discord-verify_leads_anish: Refreshing state...
forgejo_team_member.discord-verify_leads_maybe_yiyi: Refreshing state...
forgejo_team_repository.lost-and-found_lost_and_found_v2: Refreshing state...
forgejo_team_member.lost-and-found_anish: Refreshing state...
forgejo_team_member.quest_jw8: Refreshing state...
forgejo_team_repository.quest_quest: Refreshing state...
forgejo_team_member.quest_anish: Refreshing state...
forgejo_team_member.quest_kechezon: Refreshing state...
forgejo_team_member.quest_brightz: Refreshing state...
forgejo_team_member.quest_xboxbedrock: Refreshing state...
forgejo_team_member.occ-main_leads_dcrodrig: Refreshing state...
forgejo_team_member.tartan-vote_riztofu: Refreshing state...
forgejo_team_member.occ-main_leads_anish: Refreshing state...
forgejo_team_member.tartan-vote_anish: Refreshing state...
forgejo_team_repository.tartan-vote_tartan_vote: Refreshing state...
forgejo_team_member.tartan-vote_tentype: Refreshing state...
forgejo_team_member.tartan-vote_maybe_yiyi: Refreshing state...
forgejo_team_member.tartan-vote_xboxbedrock: Refreshing state...
forgejo_team_member.documentation_leads_thesuperrl: Refreshing state...
forgejo_team_member.cal_leads_anish: Refreshing state...
forgejo_team_member.documentation_leads_anish: Refreshing state...
forgejo_team_member.cal_leads_jalenluorion: Refreshing state...
forgejo_team_repository.terrier_terrier: Refreshing state...
forgejo_team_member.cmu-housing_gostmeaper: Refreshing state...
forgejo_team_member.terrier_kritdass: Refreshing state...
forgejo_team_member.cmu-housing_jessica_lr_song: Refreshing state...
forgejo_team_member.cmu-housing_manrajs: Refreshing state...
forgejo_team_member.cmu-housing_anish: Refreshing state...
forgejo_team_member.cmu-housing_nikhilr: Refreshing state...
forgejo_team_repository.cmu-housing_housing: Refreshing state...
forgejo_team_member.cmu-housing_emmawong: Refreshing state...
forgejo_team_member.cmu-housing_xboxbedrock: Refreshing state...
forgejo_team_member.cmu-housing_rhsen: Refreshing state...
forgejo_team_member.cmu-housing_stringatlas: Refreshing state...
forgejo_team_member.cmu-housing_amzoeee: Refreshing state...
forgejo_team_member.terrier_etashjha: Refreshing state...
forgejo_team_member.terrier_anish: Refreshing state...
forgejo_team_member.terrier_kothurvidur: Refreshing state...
forgejo_team_member.terrier_amzoeee: Refreshing state...
forgejo_team_member.terrier_dkou: Refreshing state...
forgejo_team_member.cmu-housing_kluoxx: Refreshing state...
forgejo_team_member.cmu-housing_tentype: Refreshing state...
forgejo_team_member.cmugpt-mcp-server_leads_anish: Refreshing state...
forgejo_team_member.cmugpt-mcp-server_leads_krishsax: Refreshing state...
forgejo_team_member.cmu-housing_zoexue: Refreshing state...
forgejo_team_member.cmugpt-agent_leads_anish: Refreshing state...
forgejo_team_member.cmugpt-agent_leads_krishsax: Refreshing state...
forgejo_team_member.groupme-mirror_leads_anish: Refreshing state...
forgejo_team_member.groupme-mirror_leads_tentype: Refreshing state...
forgejo_team_member.groupme-mirror_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.cmu-courses_leads_etashjha: Refreshing state...
forgejo_team_member.cmu-courses_leads_anish: Refreshing state...
forgejo_team_repository.cmu-insta_insta: Refreshing state...
forgejo_team_member.cmu-insta_etashjha: Refreshing state...
forgejo_team_member.components_leads_anish: Refreshing state...
forgejo_team_member.cbp_liamwalk: Refreshing state...
forgejo_team_member.cbp_xboxbedrock: Refreshing state...
forgejo_team_member.cbp_nikhilr: Refreshing state...
forgejo_team_repository.cbp_bus_sign: Refreshing state...
forgejo_team_repository.cbp_discord_verify: Refreshing state...
forgejo_team_repository.cbp_groupme_mirror: Refreshing state...
forgejo_team_repository.cbp_dalmatian: Refreshing state...
forgejo_team_member.cbp_maybe_yiyi: Refreshing state...
forgejo_team_member.cbp_tentype: Refreshing state...
forgejo_team_member.cbp_anish: Refreshing state...
forgejo_team_member.housing_leads_tentype: Refreshing state...
forgejo_team_member.housing_leads_nikhilr: Refreshing state...
forgejo_team_member.housing_leads_anish: Refreshing state...
forgejo_team_member.housing_leads_emmawong: Refreshing state...
forgejo_team_member.housing_leads_gostmeaper: Refreshing state...
forgejo_team_member.devops_leads_thesuperrl: Refreshing state...
forgejo_team_member.devops_leads_anish: Refreshing state...
forgejo_team_member.uia_anish: Refreshing state...
forgejo_team_repository.uia_components: Refreshing state...
forgejo_team_repository.uia_scottylabs_org: Refreshing state...
forgejo_team_member.terrier_leads_kritdass: Refreshing state...
forgejo_team_member.terrier_leads_anish: Refreshing state...
forgejo_team_member.slai_xboxbedrock: Refreshing state...
forgejo_team_member.slai_qianxuege: Refreshing state...
forgejo_team_repository.slai_mcp_server: Refreshing state...
forgejo_team_member.slai_krishsax: Refreshing state...
forgejo_team_member.slai_jc_965: Refreshing state...
forgejo_team_member.slai_jiyas: Refreshing state...
forgejo_team_member.slai_rzhu3: Refreshing state...
forgejo_team_repository.slai_cmugpt_agent: Refreshing state...
forgejo_team_repository.slai_cmugpt_surface: Refreshing state...
forgejo_team_repository.slai_cmugpt_sms_surface: Refreshing state...
forgejo_team_member.slai_anish: Refreshing state...
forgejo_team_member.quest_leads_anish: Refreshing state...
forgejo_team_member.quest_leads_xboxbedrock: Refreshing state...
forgejo_team_member.maps_leads_anish: Refreshing state...
forgejo_team_member.maps_leads_siddharthr: Refreshing state...
forgejo_team_repository.cmu-maps_maps: Refreshing state...
forgejo_team_member.cmu-maps_siddharthr: Refreshing state...
forgejo_team_member.cmu-maps_anish: Refreshing state...
forgejo_team_member.eats_leads_arom1a: Refreshing state...
forgejo_team_member.occ_acchs: Refreshing state...
forgejo_team_member.occ_xboxbedrock: Refreshing state...
forgejo_team_repository.occ_occ: Refreshing state...
forgejo_team_member.occ_anish: Refreshing state...
forgejo_team_member.occ_dcrodrig: Refreshing state...
forgejo_team_repository.cmu-eats_eats: Refreshing state...
forgejo_team_member.cmu-eats_arom1a: Refreshing state...
github_repository_deploy_key.occ_occ_mirror_deploy_key: Refreshing state... [id=occ:161720242]
github_repository_deploy_key.slai_cmugpt_sms_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-sms-surface:159665365]
forgejo_team_repository.cmu-eats_dining_api: Refreshing state...
github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Refreshing state... [id=dining-api:161772815]
github_repository_deploy_key.devops_documentation_mirror_deploy_key: Refreshing state... [id=documentation:159665362]
github_repository_deploy_key.slai_cmugpt_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-surface:159665348]
github_repository_deploy_key.terrier_terrier_mirror_deploy_key: Refreshing state... [id=terrier:159665339]
github_repository_deploy_key.cmu-insta_insta_mirror_deploy_key: Refreshing state... [id=insta:161720245]
github_repository_deploy_key.cbp_discord_verify_mirror_deploy_key: Refreshing state... [id=discord-verify:159665340]
github_repository_deploy_key.slai_cmugpt_agent_mirror_deploy_key: Refreshing state... [id=cmugpt-agent:159665360]
github_repository_deploy_key.devops_governance_mirror_deploy_key: Refreshing state... [id=governance:159665346]
github_repository_deploy_key.cmu-housing_housing_mirror_deploy_key: Refreshing state... [id=housing:159665369]
github_repository_deploy_key.uia_components_mirror_deploy_key: Refreshing state... [id=components:159665353]
github_repository_deploy_key.cmu-maps_maps_mirror_deploy_key: Refreshing state... [id=maps:159665325]
github_repository_ruleset.devops_infrastructure_ruleset: Refreshing state... [id=11275118]
github_repository_ruleset.devops_collie_ruleset: Refreshing state... [id=18865204]
github_repository_ruleset.quest_quest_ruleset: Refreshing state... [id=15750052]
github_repository_ruleset.cmu-cal_cal_ruleset: Refreshing state... [id=18919281]
github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Refreshing state... [id=16848966]
github_repository_ruleset.uia_scottylabs_org_ruleset: Refreshing state... [id=18112214]
github_repository_ruleset.cmu-courses_internet_archive_ruleset: Refreshing state... [id=11988074]
forgejo_team_repository.link-shortener_leads_link_shortener: Refreshing state...
forgejo_team_repository.devops_leads_kennel: Refreshing state...
forgejo_team_repository.devops_leads_governance: Refreshing state...
forgejo_team_repository.devops_leads_infrastructure: Refreshing state...
forgejo_team_repository.devops_leads_collie: Refreshing state...
forgejo_team_repository.documentation_leads_documentation: Refreshing state...
github_repository_ruleset.devops_link_shortener_ruleset: Refreshing state... [id=17911647]
forgejo_team_repository.cmu-courses_leads_internet_archive: Refreshing state...
forgejo_team_repository.courses_leads_courses: Refreshing state...
forgejo_team_repository.cal_leads_cal: Refreshing state...
github_repository_ruleset.devops_kennel_ruleset: Refreshing state... [id=13193453]
forgejo_team_repository.lost-and-found-v2_leads_lost_and_found_v2: Refreshing state...
forgejo_team_repository.quest_leads_quest: Refreshing state...
forgejo_team_repository.tartan-vote_leads_tartan_vote: Refreshing state...
github_repository_ruleset.slai_mcp_server_ruleset: Refreshing state... [id=17082154]
forgejo_team_repository.terrier_leads_terrier: Refreshing state...
forgejo_team_repository.housing_leads_housing: Refreshing state...
github_repository_ruleset.cbp_dalmatian_ruleset: Refreshing state... [id=17238915]
forgejo_team_repository.insta_leads_insta: Refreshing state...
forgejo_team_repository.discord-verify_leads_discord_verify: Refreshing state...
forgejo_team_repository.groupme-mirror_leads_groupme_mirror: Refreshing state...
forgejo_team_repository.bus-sign_leads_bus_sign: Refreshing state...
forgejo_team_repository.dalmatian_leads_dalmatian: Refreshing state...
github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Refreshing state... [id=18920618]
forgejo_team_repository.components_leads_components: Refreshing state...
forgejo_team_repository.scottylabs-org_leads_scottylabs_org: Refreshing state...
forgejo_team_repository.cmugpt-mcp-server_leads_mcp_server: Refreshing state...
forgejo_team_repository.cmugpt-surface_leads_cmugpt_surface: Refreshing state...
forgejo_team_repository.cmugpt-agent_leads_cmugpt_agent: Refreshing state...
forgejo_team_repository.cmugpt-sms-surface_leads_cmugpt_sms_surface: Refreshing state...
github_repository_ruleset.cmu-courses_courses_ruleset: Refreshing state... [id=15855161]
forgejo_team_repository.maps_leads_maps: Refreshing state...
forgejo_team_repository.occ-main_leads_occ: Refreshing state...
github_repository_ruleset.cbp_groupme_mirror_ruleset: Refreshing state... [id=12503296]
github_repository_ruleset.cmu-eats_eats_ruleset: Refreshing state... [id=21330218]
github_repository_ruleset.cbp_bus_sign_ruleset: Refreshing state... [id=15236550]
github_repository_ruleset.occ_occ_ruleset: Refreshing state... [id=21841634]
github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Refreshing state... [id=17653622]
github_repository_ruleset.cmu-eats_dining_api_ruleset: Refreshing state... [id=21879626]
github_repository_ruleset.devops_documentation_ruleset: Refreshing state... [id=17079368]
github_repository_ruleset.terrier_terrier_ruleset: Refreshing state... [id=12671003]
github_repository_ruleset.slai_cmugpt_surface_ruleset: Refreshing state... [id=16476662]
github_repository_ruleset.cmu-insta_insta_ruleset: Refreshing state... [id=21841636]
github_repository_ruleset.cbp_discord_verify_ruleset: Refreshing state... [id=11449154]
github_repository_ruleset.slai_cmugpt_agent_ruleset: Refreshing state... [id=16476661]
github_repository_ruleset.devops_governance_ruleset: Refreshing state... [id=14488792]
github_repository_ruleset.cmu-housing_housing_ruleset: Refreshing state... [id=17421604]
github_repository_ruleset.uia_components_ruleset: Refreshing state... [id=16366085]
github_repository_ruleset.cmu-maps_maps_ruleset: Refreshing state... [id=18864961]

Note: Objects have changed outside of OpenTofu

OpenTofu detected the following changes made outside of OpenTofu since the
last "tofu apply" which may have affected this plan:

  # restapi_object.cmu-eats_dining_api_push_mirror has been deleted
  - resource "restapi_object" "cmu-eats_dining_api_push_mirror" {
      - api_data        = {
          - "public_key"     = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIIKyjdH3wH8J3aSgeZMLB9/uUGzyYltQsk5Cl1Q8MXqF"
            # (9 unchanged elements hidden)
        } -> null
        id              = "remote_mirror_a4MsQ5rO1_Z"
        # (5 unchanged attributes hidden)
    }

  # restapi_object.cmu-eats_eats_push_mirror has been deleted
  - resource "restapi_object" "cmu-eats_eats_push_mirror" {
      - api_data        = {
          - "public_key"     = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFsm79Zlyi0JKH4y/BdgCCoLyrC4/18yF3owfsqOt53/"
            # (9 unchanged elements hidden)
        } -> null
        id              = "remote_mirror_sjYwMA6NPgh"
        # (5 unchanged attributes hidden)
    }


Unless you have made equivalent changes to your configuration, or ignored the
relevant attributes using ignore_changes, the following plan may include
actions to undo or respond to these changes.

─────────────────────────────────────────────────────────────────────────────

OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
  + create
  ~ update in-place (current -> planned)
-/+ destroy and then create replacement

OpenTofu planned the following actions, but then encountered a problem:

  # forgejo_repository.cmu-eats_dining_api will be updated in-place
  ~ resource "forgejo_repository" "cmu-eats_dining_api" {
      ~ clone_url                   = "https://git.cmu.dev/ScottyLabs/dining-api.git" -> (known after apply)
      ~ created_at                  = "2026-09-18T23:16:27Z" -> (known after apply)
      + description                 = "Dining API for CMU dining locations and availability"
      ~ empty                       = false -> (known after apply)
      + external_tracker            = (known after apply)
      + external_wiki               = (known after apply)
      ~ forks_count                 = 0 -> (known after apply)
      ~ full_name                   = "ScottyLabs/dining-api" -> (known after apply)
      ~ html_url                    = "https://git.cmu.dev/ScottyLabs/dining-api" -> (known after apply)
        id                          = 97
      ~ internal_tracker            = {
          ~ allow_only_contributors_to_track_time = true -> (known after apply)
          ~ enable_issue_dependencies             = true -> (known after apply)
          ~ enable_time_tracker                   = true -> (known after apply)
        } -> (known after apply)
      + mirror_interval             = (known after apply)
      ~ mirror_updated              = "0001-01-01T00:00:00Z" -> (known after apply)
        name                        = "dining-api"
      ~ open_issues_count           = 0 -> (known after apply)
      ~ open_pr_counter             = 5 -> (known after apply)
      ~ permissions                 = {
          ~ admin = true -> (known after apply)
          ~ pull  = true -> (known after apply)
          ~ push  = true -> (known after apply)
        } -> (known after apply)
      ~ release_counter             = 0 -> (known after apply)
      ~ size                        = 38816 -> (known after apply)
      ~ ssh_url                     = "ssh://forgejo@git.cmu.dev/ScottyLabs/dining-api.git" -> (known after apply)
      ~ stars_count                 = 0 -> (known after apply)
      ~ updated_at                  = "2026-09-18T23:22:22Z" -> (known after apply)
      ~ watchers_count              = 3 -> (known after apply)
      ~ website                     = "https://git.cmu.dev/ScottyLabs/dining-api" -> "https://api.cmueats.com"
        # (30 unchanged attributes hidden)
    }

  # forgejo_repository.cmu-eats_eats will be updated in-place
  ~ resource "forgejo_repository" "cmu-eats_eats" {
      ~ clone_url                   = "https://git.cmu.dev/ScottyLabs/eats.git" -> (known after apply)
      ~ created_at                  = "2026-09-18T23:15:39Z" -> (known after apply)
      ~ description                 = "Vite web app for CMU dining location info" -> "CMU dining location info"
      ~ empty                       = false -> (known after apply)
      + external_tracker            = (known after apply)
      + external_wiki               = (known after apply)
      ~ forks_count                 = 0 -> (known after apply)
      ~ full_name                   = "ScottyLabs/eats" -> (known after apply)
      ~ html_url                    = "https://git.cmu.dev/ScottyLabs/eats" -> (known after apply)
        id                          = 96
      ~ internal_tracker            = {
          ~ allow_only_contributors_to_track_time = true -> (known after apply)
          ~ enable_issue_dependencies             = true -> (known after apply)
          ~ enable_time_tracker                   = true -> (known after apply)
        } -> (known after apply)
      + mirror_interval             = (known after apply)
      ~ mirror_updated              = "0001-01-01T00:00:00Z" -> (known after apply)
        name                        = "eats"
      ~ open_issues_count           = 48 -> (known after apply)
      ~ open_pr_counter             = 4 -> (known after apply)
      ~ permissions                 = {
          ~ admin = true -> (known after apply)
          ~ pull  = true -> (known after apply)
          ~ push  = true -> (known after apply)
        } -> (known after apply)
      ~ release_counter             = 0 -> (known after apply)
      ~ size                        = 95496 -> (known after apply)
      ~ ssh_url                     = "ssh://forgejo@git.cmu.dev/ScottyLabs/eats.git" -> (known after apply)
      ~ stars_count                 = 0 -> (known after apply)
      ~ updated_at                  = "2026-09-18T23:35:12Z" -> (known after apply)
      ~ watchers_count              = 3 -> (known after apply)
        # (31 unchanged attributes hidden)
    }

  # forgejo_repository.cmu-study_study will be created
  + resource "forgejo_repository" "cmu-study_study" {
      + allow_manual_merge          = false
      + allow_merge_commits         = true
      + allow_rebase                = true
      + allow_rebase_explicit       = true
      + allow_squash_merge          = true
      + archive_on_destroy          = false
      + archived                    = false
      + auto_init                   = true
      + autodetect_manual_merge     = false
      + avatar_url                  = (known after apply)
      + clone_url                   = (known after apply)
      + created_at                  = (known after apply)
      + default_branch              = "main"
      + default_merge_style         = "merge"
      + description                 = "A platform for creating and joining study groups"
      + empty                       = (known after apply)
      + external_tracker            = (known after apply)
      + external_wiki               = (known after apply)
      + fork                        = (known after apply)
      + forks_count                 = (known after apply)
      + full_name                   = (known after apply)
      + has_actions                 = true
      + has_issues                  = true
      + has_packages                = true
      + has_projects                = true
      + has_pull_requests           = true
      + has_releases                = true
      + has_wiki                    = true
      + html_url                    = (known after apply)
      + id                          = (known after apply)
      + ignore_whitespace_conflicts = false
      + internal                    = (known after apply)
      + internal_tracker            = (known after apply)
      + labels                      = false
      + lfs                         = false
      + milestones                  = false
      + mirror                      = false
      + mirror_interval             = (known after apply)
      + mirror_updated              = (known after apply)
      + name                        = "study"
      + open_issues_count           = (known after apply)
      + open_pr_counter             = (known after apply)
      + owner                       = "ScottyLabs"
      + parent_id                   = (known after apply)
      + permissions                 = (known after apply)
      + private                     = false
      + release_counter             = (known after apply)
      + size                        = (known after apply)
      + ssh_url                     = (known after apply)
      + stars_count                 = (known after apply)
      + template                    = false
      + trust_model                 = "default"
      + updated_at                  = (known after apply)
      + watchers_count              = (known after apply)
      + website                     = "https://cmustudy.com"
    }

  # forgejo_team.cmu-study will be created
  + resource "forgejo_team" "cmu-study" {
      + can_create_org_repo       = (known after apply)
      + description               = "CMU Study"
      + id                        = (known after apply)
      + includes_all_repositories = (known after apply)
      + name                      = "cmu-study"
      + organization              = (known after apply)
      + organization_id           = 3
      + permission                = "write"
      + units_map                 = {
          + "repo.code"   = "write"
          + "repo.issues" = "write"
          + "repo.pulls"  = "write"
        }
    }

  # forgejo_team.study_leads will be created
  + resource "forgejo_team" "study_leads" {
      + can_create_org_repo       = false
      + description               = "CMU Study leads"
      + id                        = (known after apply)
      + includes_all_repositories = false
      + name                      = "study-leads"
      + organization              = (known after apply)
      + organization_id           = 3
      + permission                = "admin"
      + units_map                 = {
          + "repo.actions"    = "admin"
          + "repo.code"       = "admin"
          + "repo.ext_issues" = "admin"
          + "repo.ext_wiki"   = "admin"
          + "repo.issues"     = "admin"
          + "repo.packages"   = "admin"
          + "repo.projects"   = "admin"
          + "repo.pulls"      = "admin"
          + "repo.releases"   = "admin"
          + "repo.wiki"       = "admin"
        }
    }

  # forgejo_team_member.cmu-study_annadavi will be created
  + resource "forgejo_team_member" "cmu-study_annadavi" {
      + team_id = (known after apply)
      + user    = "annadavi"
    }

  # forgejo_team_member.study_leads_annadavi will be created
  + resource "forgejo_team_member" "study_leads_annadavi" {
      + team_id = (known after apply)
      + user    = "annadavi"
    }

  # forgejo_team_repository.cmu-study_study will be created
  + resource "forgejo_team_repository" "cmu-study_study" {
      + owner      = "ScottyLabs"
      + repository = "study"
      + team_id    = (known after apply)
    }

  # forgejo_team_repository.study_leads_study will be created
  + resource "forgejo_team_repository" "study_leads_study" {
      + owner      = "ScottyLabs"
      + repository = "study"
      + team_id    = (known after apply)
    }

  # github_actions_repository_permissions.cmu-study_study_actions will be created
  + resource "github_actions_repository_permissions" "cmu-study_study_actions" {
      + enabled              = false
      + id                   = (known after apply)
      + repository           = "study"
      + sha_pinning_required = (known after apply)
    }

  # github_repository.cmu-study_study will be created
  + resource "github_repository" "cmu-study_study" {
      + allow_auto_merge                        = false
      + allow_forking                           = (known after apply)
      + allow_merge_commit                      = true
      + allow_rebase_merge                      = true
      + allow_squash_merge                      = true
      + archived                                = false
      + default_branch                          = (known after apply)
      + delete_branch_on_merge                  = false
      + description                             = "READ-ONLY MIRROR"
      + etag                                    = (known after apply)
      + fork                                    = (known after apply)
      + full_name                               = (known after apply)
      + git_clone_url                           = (known after apply)
      + has_discussions                         = false
      + has_issues                              = false
      + has_projects                            = false
      + has_wiki                                = false
      + homepage_url                            = "https://git.cmu.dev/ScottyLabs/study"
      + html_url                                = (known after apply)
      + http_clone_url                          = (known after apply)
      + id                                      = (known after apply)
      + ignore_vulnerability_alerts_during_read = false
      + merge_commit_message                    = "PR_TITLE"
      + merge_commit_title                      = "MERGE_MESSAGE"
      + name                                    = "study"
      + node_id                                 = (known after apply)
      + primary_language                        = (known after apply)
      + private                                 = (known after apply)
      + repo_id                                 = (known after apply)
      + source_owner                            = (known after apply)
      + source_repo                             = (known after apply)
      + squash_merge_commit_message             = "COMMIT_MESSAGES"
      + squash_merge_commit_title               = "COMMIT_OR_PR_TITLE"
      + ssh_clone_url                           = (known after apply)
      + svn_url                                 = (known after apply)
      + topics                                  = (known after apply)
      + visibility                              = "public"
      + vulnerability_alerts                    = (known after apply)
      + web_commit_signoff_required             = (known after apply)

      + security_and_analysis (known after apply)
    }

  # github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key must be replaced
-/+ resource "github_repository_deploy_key" "cmu-eats_dining_api_mirror_deploy_key" {
      ~ etag       = "W/\"bf394a0fcc9e1231b4ee544598e6680359d71d177764d9f0c8cd22f5c7ec3ef4\"" -> (known after apply)
      ~ id         = "dining-api:161772815" -> (known after apply)
      ~ key        = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIIKyjdH3wH8J3aSgeZMLB9/uUGzyYltQsk5Cl1Q8MXqF" # forces replacement -> (known after apply) # forces replacement
        # (3 unchanged attributes hidden)
    }

  # github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key must be replaced
-/+ resource "github_repository_deploy_key" "cmu-eats_eats_mirror_deploy_key" {
      ~ etag       = "W/\"5cf445e3b52f1b2e59fe1cb32c5c2e0859e2598566a7aa75c3ebdbf4b01f6164\"" -> (known after apply)
      ~ id         = "eats:161775995" -> (known after apply)
      ~ key        = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFsm79Zlyi0JKH4y/BdgCCoLyrC4/18yF3owfsqOt53/" # forces replacement -> (known after apply) # forces replacement
        # (3 unchanged attributes hidden)
    }

  # github_repository_deploy_key.cmu-study_study_mirror_deploy_key will be created
  + resource "github_repository_deploy_key" "cmu-study_study_mirror_deploy_key" {
      + etag       = (known after apply)
      + id         = (known after apply)
      + key        = (known after apply)
      + read_only  = false
      + repository = "study"
      + title      = "Forgejo mirror"
    }

  # github_repository_ruleset.cmu-eats_dining_api_ruleset will be updated in-place
  ~ resource "github_repository_ruleset" "cmu-eats_dining_api_ruleset" {
        id          = "21879626"
        name        = "Default"
        # (6 unchanged attributes hidden)

      ~ bypass_actors {
          ~ actor_id    = 161772815 -> (known after apply)
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place
  ~ resource "github_repository_ruleset" "cmu-eats_eats_ruleset" {
        id          = "21330218"
        name        = "Default"
        # (6 unchanged attributes hidden)

      ~ bypass_actors {
          ~ actor_id    = 161775995 -> (known after apply)
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.cmu-study_study_ruleset will be created
  + resource "github_repository_ruleset" "cmu-study_study_ruleset" {
      + enforcement = "active"
      + etag        = (known after apply)
      + id          = (known after apply)
      + name        = "Default"
      + node_id     = (known after apply)
      + repository  = "study"
      + ruleset_id  = (known after apply)
      + target      = "branch"

      + bypass_actors {
          + actor_id    = (known after apply)
          + actor_type  = "DeployKey"
          + bypass_mode = "always"
        }

      + conditions {
          + ref_name {
              + exclude = []
              + include = [
                  + "~ALL",
                ]
            }
        }

      + rules {
          + creation                      = true
          + deletion                      = true
          + non_fast_forward              = true
          + update                        = true
          + update_allows_fetch_and_merge = false

          + pull_request {
              + allowed_merge_methods             = (known after apply)
              + dismiss_stale_reviews_on_push     = false
              + require_code_owner_review         = false
              + require_last_push_approval        = false
              + required_approving_review_count   = 1
              + required_review_thread_resolution = false
            }
        }
    }

  # github_team.cmu-study will be created
  + resource "github_team" "cmu-study" {
      + create_default_maintainer = false
      + etag                      = (known after apply)
      + id                        = (known after apply)
      + members_count             = (known after apply)
      + name                      = "CMU Study"
      + node_id                   = (known after apply)
      + notification_setting      = "notifications_enabled"
      + parent_team_read_id       = (known after apply)
      + parent_team_read_slug     = (known after apply)
      + privacy                   = "closed"
      + slug                      = (known after apply)
    }

  # github_team_membership.cmu-study_annadavi will be created
  + resource "github_team_membership" "cmu-study_annadavi" {
      + etag     = (known after apply)
      + id       = (known after apply)
      + role     = "maintainer"
      + team_id  = (known after apply)
      + username = "annadavid1"
    }

  # restapi_object.cmu-eats_dining_api_push_mirror will be created
  + resource "restapi_object" "cmu-eats_dining_api_push_mirror" {
      + api_data        = (known after apply)
      + api_response    = (known after apply)
      + create_response = (known after apply)
      + data            = jsonencode(
            {
              + interval       = "8h0m0s"
              + remote_address = "git@github.com:ScottyLabs/dining-api.git"
              + sync_on_commit = true
              + use_ssh        = true
            }
        )
      + id              = (known after apply)
      + id_attribute    = "remote_name"
      + path            = "/api/v1/repos/ScottyLabs/dining-api/push_mirrors"
    }

  # restapi_object.cmu-eats_eats_kennel_webhook will be created
  + resource "restapi_object" "cmu-eats_eats_kennel_webhook" {
      + api_data        = (known after apply)
      + api_response    = (known after apply)
      + create_response = (known after apply)
      + data            = (sensitive value)
      + id              = (known after apply)
      + id_attribute    = "id"
      + path            = "/api/v1/repos/ScottyLabs/eats/hooks"
      + update_method   = "PATCH"
    }

  # restapi_object.cmu-eats_eats_push_mirror will be created
  + resource "restapi_object" "cmu-eats_eats_push_mirror" {
      + api_data        = (known after apply)
      + api_response    = (known after apply)
      + create_response = (known after apply)
      + data            = jsonencode(
            {
              + interval       = "8h0m0s"
              + remote_address = "git@github.com:ScottyLabs/eats.git"
              + sync_on_commit = true
              + use_ssh        = true
            }
        )
      + id              = (known after apply)
      + id_attribute    = "remote_name"
      + path            = "/api/v1/repos/ScottyLabs/eats/push_mirrors"
    }

  # restapi_object.cmu-study_study_kennel_webhook will be created
  + resource "restapi_object" "cmu-study_study_kennel_webhook" {
      + api_data        = (known after apply)
      + api_response    = (known after apply)
      + create_response = (known after apply)
      + data            = (sensitive value)
      + id              = (known after apply)
      + id_attribute    = "id"
      + path            = "/api/v1/repos/ScottyLabs/study/hooks"
      + update_method   = "PATCH"
    }

  # restapi_object.cmu-study_study_push_mirror will be created
  + resource "restapi_object" "cmu-study_study_push_mirror" {
      + api_data        = (known after apply)
      + api_response    = (known after apply)
      + create_response = (known after apply)
      + data            = jsonencode(
            {
              + interval       = "8h0m0s"
              + remote_address = "git@github.com:ScottyLabs/study.git"
              + sync_on_commit = true
              + use_ssh        = true
            }
        )
      + id              = (known after apply)
      + id_attribute    = "remote_name"
      + path            = "/api/v1/repos/ScottyLabs/study/push_mirrors"
    }

Plan: 20 to add, 4 to change, 2 to destroy.
╷
│ Error: Unable to read team repository
│ 
│   with forgejo_team_repository.cmu-eats_dining_api,
│   on forgejo_team_repos.tf.json line 102, in resource.forgejo_team_repository.cmu-eats_dining_api:
│  102:       },
│ 
│ Repository 'ScottyLabs/dining-api' not found in team with ID 28
╵
╷
│ Error: Unable to read team repository
│ 
│   with forgejo_team_repository.cmu-eats_eats,
│   on forgejo_team_repos.tf.json line 111, in resource.forgejo_team_repository.cmu-eats_eats:
│  111:       },
│ 
│ Repository 'ScottyLabs/eats' not found in team with ID 28
╵



Plan Summary

10 projects, 8 with changes, 1 with no changes, 1 failed

  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 10 projects: 1. project: `posthog` dir: `tofu/posthog` workspace: `default` 1. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `openbao` dir: `tofu/openbao` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `sentry` dir: `tofu/sentry` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `posthog` dir: `tofu/posthog` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.10s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # posthog_project.cmu-study_study will be created + resource "posthog_project" "cmu-study_study" { + api_token = (sensitive value) + id = (known after apply) + name = "study" + organization_id = (known after apply) + timezone = (known after apply) } # vault_kv_secret_v2.cmu-study_study_posthog_host will be created + resource "vault_kv_secret_v2" "cmu-study_study_posthog_host" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/POSTHOG_HOST" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.cmu-study_study_posthog_key will be created + resource "vault_kv_secret_v2" "cmu-study_study_posthog_key" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/POSTHOG_KEY" + path = (known after apply) + custom_metadata (known after apply) } Plan: 3 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p posthog ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fposthog%252Fdefault%252Fposthog) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p posthog ``` Plan: 3 to add, 0 to change, 0 to destroy. --- ### 2. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.25s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_bridge_login_cmu_study will be created + resource "null_resource" "slack_bridge_login_cmu_study" { + id = (known after apply) + triggers = { + "channel" = "C09KN6M3P53" } } # synapse_mautrix_slack_link.cmu_study will be created + resource "synapse_mautrix_slack_link" "cmu_study" { + discord_channel_id = "1512270354943578132" + discord_relay_configured = (known after apply) + id = (known after apply) + room_alias = (known after apply) + room_id = (known after apply) + slack_bridged = (known after apply) + slack_channel_id = "C09KN6M3P53" + team_name = "CMU Study" + team_slug = "cmu-study" } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p matrix_bridges ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fmatrix_bridges%252Fdefault%252Fmatrix_bridges) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p matrix_bridges ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 3. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.43s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_cmu-study_ch0_annadavi will be created + resource "null_resource" "slack_cmu-study_ch0_annadavi" { + id = (known after apply) + triggers = { + "channel" = "C09KN6M3P53" + "user" = "U07M24AKRGV" } } # null_resource.slack_hub_annadavi will be created + resource "null_resource" "slack_hub_annadavi" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U07M24AKRGV" } } # null_resource.slack_leads_annadavi will be created + resource "null_resource" "slack_leads_annadavi" { + id = (known after apply) + triggers = { + "channel" = "C0AN1M0HDV0" + "user" = "U07M24AKRGV" } } Plan: 3 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 3 to add, 0 to change, 0 to destroy. --- ### 4. project: `openbao` dir: `tofu/openbao` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.25s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # vault_identity_group.bus-sign_admins will be updated in-place ! resource "vault_identity_group" "bus-sign_admins" { id = "1b72d8b5-8511-fe92-c98c-2ddb8cbf17a1" name = "bus-sign-admins" ! policies = [ - "bus-sign-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.cal_admins will be updated in-place ! resource "vault_identity_group" "cal_admins" { id = "da96ea82-8537-f0a1-0d42-a7ab1cc3c006" name = "cal-admins" ! policies = [ - "cal-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.changelog_admins will be updated in-place ! resource "vault_identity_group" "changelog_admins" { id = "d73c0ef4-a7d2-d171-ebd1-1c15cee174c8" name = "changelog-admins" ! policies = [ - "changelog-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.cmugpt-agent_admins will be updated in-place ! resource "vault_identity_group" "cmugpt-agent_admins" { id = "cb081cd2-a3cd-81e1-8c20-0914cac5c213" name = "cmugpt-agent-admins" ! policies = [ - "cmugpt-agent-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.cmugpt-mcp-server_admins will be updated in-place ! resource "vault_identity_group" "cmugpt-mcp-server_admins" { id = "a4717a6f-015f-9fae-c6a2-76657b8a7464" name = "cmugpt-mcp-server-admins" ! policies = [ - "cmugpt-mcp-server-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.cmugpt-sms-surface_admins will be updated in-place ! resource "vault_identity_group" "cmugpt-sms-surface_admins" { id = "780f9498-dee7-8cac-9bbc-e1feab5ea2fb" name = "cmugpt-sms-surface-admins" ! policies = [ - "cmugpt-sms-surface-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.cmugpt-surface_admins will be updated in-place ! resource "vault_identity_group" "cmugpt-surface_admins" { id = "3ae18ef3-f370-f163-767a-accfb092ea0d" name = "cmugpt-surface-admins" ! policies = [ - "cmugpt-surface-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.components_admins will be updated in-place ! resource "vault_identity_group" "components_admins" { id = "d2971366-64a4-984e-1e4e-dad3cd07c841" name = "components-admins" ! policies = [ - "components-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.courses_admins will be updated in-place ! resource "vault_identity_group" "courses_admins" { id = "d090ae17-1992-86a5-1f8b-092c937092e8" name = "courses-admins" ! policies = [ - "courses-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.dalmatian_admins will be updated in-place ! resource "vault_identity_group" "dalmatian_admins" { id = "b1d88b18-8b44-c2ea-9220-fc53ee7e3c7e" name = "dalmatian-admins" ! policies = [ - "dalmatian-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.discord-verify_admins will be updated in-place ! resource "vault_identity_group" "discord-verify_admins" { id = "c3e29b3b-0ce5-61a3-3a69-e950f297fa78" name = "discord-verify-admins" ! policies = [ - "discord-verify-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.documentation_admins will be updated in-place ! resource "vault_identity_group" "documentation_admins" { id = "35c5df83-0e17-31fc-7b82-3365d5526dd5" name = "documentation-admins" ! policies = [ - "documentation-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.eats_admins will be updated in-place ! resource "vault_identity_group" "eats_admins" { id = "05959815-c896-9703-32eb-7ba84604303c" name = "eats-admins" ! policies = [ - "eats-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.groupme-mirror_admins will be updated in-place ! resource "vault_identity_group" "groupme-mirror_admins" { id = "e264aabf-8766-67fd-f0bf-860a240ee2c2" name = "groupme-mirror-admins" ! policies = [ - "groupme-mirror-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.housing_admins will be updated in-place ! resource "vault_identity_group" "housing_admins" { id = "7c688f49-59f1-1593-1e19-16f70b86eae2" name = "housing-admins" ! policies = [ - "housing-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.insta_admins will be updated in-place ! resource "vault_identity_group" "insta_admins" { id = "be24e607-e8d5-9fa3-a11a-9a8ca05ede04" name = "insta-admins" ! policies = [ - "insta-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.link-shortener_admins will be updated in-place ! resource "vault_identity_group" "link-shortener_admins" { id = "d3fa8791-4bd4-db20-68b6-79db5ac36a25" name = "link-shortener-admins" ! policies = [ - "link-shortener-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.lost-and-found-v2_admins will be updated in-place ! resource "vault_identity_group" "lost-and-found-v2_admins" { id = "db8b5072-bb86-c9d0-e93f-951c2901bd15" name = "lost-and-found-v2-admins" ! policies = [ - "lost-and-found-v2-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.maps_admins will be updated in-place ! resource "vault_identity_group" "maps_admins" { id = "6b2bd85e-e311-ae83-8a07-30100f0ac12f" name = "maps-admins" ! policies = [ - "maps-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.occ-main_admins will be updated in-place ! resource "vault_identity_group" "occ-main_admins" { id = "4d40db70-1bac-505e-ec8d-5a9d5cf15ac7" name = "occ-main-admins" ! policies = [ - "occ-main-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.quest_admins will be updated in-place ! resource "vault_identity_group" "quest_admins" { id = "c03f2eab-1701-f315-fc17-39814ddcb34b" name = "quest-admins" ! policies = [ - "quest-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.scottylabs-org_admins will be updated in-place ! resource "vault_identity_group" "scottylabs-org_admins" { id = "0dbd914d-b96c-de7b-1ec5-a0ef330b20e0" name = "scottylabs-org-admins" ! policies = [ - "scottylabs-org-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.study_admins will be created + resource "vault_identity_group" "study_admins" { + external_member_entity_ids = false + external_member_group_ids = false + external_policies = false + id = (known after apply) + name = "study-admins" + policies = [ + "study-prod", ] + type = "external" } # vault_identity_group.study_members will be created + resource "vault_identity_group" "study_members" { + external_member_entity_ids = false + external_member_group_ids = false + external_policies = false + id = (known after apply) + name = "study-members" + policies = [ + "study-dev", ] + type = "external" } # vault_identity_group.tartan-vote_admins will be updated in-place ! resource "vault_identity_group" "tartan-vote_admins" { id = "ff77ad01-20b1-c320-37bd-6185b3f98d80" name = "tartan-vote-admins" ! policies = [ - "tartan-vote-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group.terrier_admins will be updated in-place ! resource "vault_identity_group" "terrier_admins" { id = "956692a1-84f6-c95c-6d40-2382ead13476" name = "terrier-admins" ! policies = [ - "terrier-dev", # (1 unchanged element hidden) ] # (7 unchanged attributes hidden) } # vault_identity_group_alias.study_admins will be created + resource "vault_identity_group_alias" "study_admins" { + canonical_id = (known after apply) + id = (known after apply) + mount_accessor = "auth_oidc_375ac979" + name = "/projects/study/admins" } # vault_identity_group_alias.study_members will be created + resource "vault_identity_group_alias" "study_members" { + canonical_id = (known after apply) + id = (known after apply) + mount_accessor = "auth_oidc_375ac979" + name = "/projects/study" } # vault_policy.study_dev will be created + resource "vault_policy" "study_dev" { + allow_overwrite = (known after apply) + id = (known after apply) + name = "study-dev" + policy = <<-EOT path "secret/data/secretspec/study/dev/*" { capabilities = ["create", "read", "update", "delete", "list"] } path "secret/metadata/secretspec/study/dev/*" { capabilities = ["list", "read", "delete"] } path "secret/data/secretspec/study/preview/*" { capabilities = ["create", "read", "update", "delete", "list"] } path "secret/metadata/secretspec/study/preview/*" { capabilities = ["list", "read", "delete"] } EOT } # vault_policy.study_prod will be created + resource "vault_policy" "study_prod" { + allow_overwrite = (known after apply) + id = (known after apply) + name = "study-prod" + policy = <<-EOT path "secret/data/secretspec/study/prod/*" { capabilities = ["create", "read", "update", "delete", "list"] } path "secret/metadata/secretspec/study/prod/*" { capabilities = ["list", "read", "delete"] } path "secret/data/secretspec/study/staging/*" { capabilities = ["create", "read", "update", "delete", "list"] } path "secret/metadata/secretspec/study/staging/*" { capabilities = ["list", "read", "delete"] } path "secret/data/secretspec/study/ci/*" { capabilities = ["create", "read", "update", "delete", "list"] } path "secret/metadata/secretspec/study/ci/*" { capabilities = ["list", "read", "delete"] } EOT } Plan: 6 to add, 24 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p openbao ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fopenbao%252Fdefault%252Fopenbao) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p openbao ``` Plan: 6 to add, 24 to change, 0 to destroy. --- ### 5. project: `google_play` dir: `tofu/google_play` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.33s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` --- ### 6. project: `sentry` dir: `tofu/sentry` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.40s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # sentry_key.cmu-study_study will be created + resource "sentry_key" "cmu-study_study" { + dsn = (sensitive value) + dsn_csp = (known after apply) + dsn_public = (known after apply) + dsn_secret = (sensitive value) + id = (known after apply) + javascript_loader_script = (known after apply) + name = "default" + organization = "scottylabs" + project = "study" + project_id = (known after apply) + public = (known after apply) + rate_limit_count = (known after apply) + rate_limit_window = (known after apply) + secret = (sensitive value) } # sentry_project.cmu-study_study will be created + resource "sentry_project" "cmu-study_study" { + client_security = (known after apply) + digests_max_delay = (known after apply) + digests_min_delay = (known after apply) + features = (known after apply) + filters = (known after apply) + fingerprinting_rules = (known after apply) + grouping_enhancements = (known after apply) + highlight_tags = (known after apply) + id = (known after apply) + internal_id = (known after apply) + name = "study" + organization = "scottylabs" + resolve_age = (known after apply) + slug = "study" + teams = [ + "cmu-study", ] } # sentry_team.cmu-study will be created + resource "sentry_team" "cmu-study" { + has_access = (known after apply) + id = (known after apply) + internal_id = (known after apply) + is_member = (known after apply) + is_pending = (known after apply) + name = "CMU Study" + organization = "scottylabs" + slug = "cmu-study" + team_id = (known after apply) } # vault_kv_secret_v2.cmu-study_study_sentry_dsn will be created + resource "vault_kv_secret_v2" "cmu-study_study_sentry_dsn" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/SENTRY_DSN" + path = (known after apply) + custom_metadata (known after apply) } Plan: 4 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p sentry ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fsentry%252Fdefault%252Fsentry) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p sentry ``` Plan: 4 to add, 0 to change, 0 to destroy. --- ### 7. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.34s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.ops_annadavi will be created + resource "google_cloud_identity_group_membership" "ops_annadavi" { + create_time = (known after apply) + group = "groups/03fwokq036zsp1f" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "annadavi@andrew.cmu.edu" } + roles { + name = "MEMBER" } } # google_cloud_identity_group_membership.tech_annadavi will be created + resource "google_cloud_identity_group_membership" "tech_annadavi" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "annadavi@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 8. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.41s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group.project_cmu-study_study will be created + resource "keycloak_group" "project_cmu-study_study" { + id = (known after apply) + name = "study" + parent_id = "9e6890c8-50a0-4b82-86b9-91791fffac25" + path = (known after apply) + realm_id = "scottylabs" } # keycloak_group.project_cmu-study_study_admins will be created + resource "keycloak_group" "project_cmu-study_study_admins" { + id = (known after apply) + name = "admins" + parent_id = (known after apply) + path = (known after apply) + realm_id = "scottylabs" } # keycloak_group_memberships.project_cmu-study_study will be created + resource "keycloak_group_memberships" "project_cmu-study_study" { + group_id = (known after apply) + id = (known after apply) + members = [ + "annadavi", ] + realm_id = "scottylabs" } # keycloak_group_memberships.project_cmu-study_study_admins will be created + resource "keycloak_group_memberships" "project_cmu-study_study_admins" { + group_id = (known after apply) + id = (known after apply) + members = [ + "annadavi", ] + realm_id = "scottylabs" } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "annadavi", # (35 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_openid_client.study will be created + resource "keycloak_openid_client" "study" { + access_token_lifespan = (known after apply) + access_type = "CONFIDENTIAL" + always_display_in_console = false + backchannel_logout_session_required = true + client_authenticator_type = "client-secret" + client_id = "study" + client_offline_session_idle_timeout = (known after apply) + client_offline_session_max_lifespan = (known after apply) + client_secret = (sensitive value) + client_secret_wo = (write-only attribute) + client_session_idle_timeout = (known after apply) + client_session_max_lifespan = (known after apply) + consent_required = (known after apply) + direct_access_grants_enabled = false + display_on_consent_screen = (known after apply) + enabled = true + exclude_issuer_from_auth_response = (known after apply) + exclude_session_state_from_auth_response = (known after apply) + frontchannel_logout_enabled = (known after apply) + full_scope_allowed = true + id = (known after apply) + implicit_flow_enabled = (known after apply) + import = false + name = "study" + oauth2_device_authorization_grant_enabled = false + oauth2_jwt_authorization_grant_enabled = false + realm_id = "scottylabs" + require_dpop_bound_tokens = (known after apply) + resource_server_id = (known after apply) + service_account_user_id = (known after apply) + service_accounts_enabled = (known after apply) + standard_flow_enabled = true + standard_token_exchange_enabled = false + use_refresh_tokens = true + use_refresh_tokens_client_credentials = false + valid_post_logout_redirect_uris = (known after apply) + valid_redirect_uris = [ + "https://oauth.scottylabs.org/oauth2/callback", ] + web_origins = [ + "+", ] } # keycloak_openid_client.study_dev will be created + resource "keycloak_openid_client" "study_dev" { + access_token_lifespan = (known after apply) + access_type = "CONFIDENTIAL" + always_display_in_console = false + backchannel_logout_session_required = true + client_authenticator_type = "client-secret" + client_id = "study-dev" + client_offline_session_idle_timeout = (known after apply) + client_offline_session_max_lifespan = (known after apply) + client_secret = (sensitive value) + client_secret_wo = (write-only attribute) + client_session_idle_timeout = (known after apply) + client_session_max_lifespan = (known after apply) + consent_required = (known after apply) + direct_access_grants_enabled = false + display_on_consent_screen = (known after apply) + enabled = true + exclude_issuer_from_auth_response = (known after apply) + exclude_session_state_from_auth_response = (known after apply) + frontchannel_logout_enabled = (known after apply) + full_scope_allowed = true + id = (known after apply) + implicit_flow_enabled = (known after apply) + import = false + name = "study-dev" + oauth2_device_authorization_grant_enabled = false + oauth2_jwt_authorization_grant_enabled = false + realm_id = "scottylabs" + require_dpop_bound_tokens = (known after apply) + resource_server_id = (known after apply) + service_account_user_id = (known after apply) + service_accounts_enabled = (known after apply) + standard_flow_enabled = true + standard_token_exchange_enabled = false + use_refresh_tokens = true + use_refresh_tokens_client_credentials = false + valid_post_logout_redirect_uris = (known after apply) + valid_redirect_uris = [ + "http://localhost:8090/oauth2/callback", ] + web_origins = [ + "+", ] } # keycloak_openid_client.study_staging will be created + resource "keycloak_openid_client" "study_staging" { + access_token_lifespan = (known after apply) + access_type = "CONFIDENTIAL" + always_display_in_console = false + backchannel_logout_session_required = true + client_authenticator_type = "client-secret" + client_id = "study-staging" + client_offline_session_idle_timeout = (known after apply) + client_offline_session_max_lifespan = (known after apply) + client_secret = (sensitive value) + client_secret_wo = (write-only attribute) + client_session_idle_timeout = (known after apply) + client_session_max_lifespan = (known after apply) + consent_required = (known after apply) + direct_access_grants_enabled = false + display_on_consent_screen = (known after apply) + enabled = true + exclude_issuer_from_auth_response = (known after apply) + exclude_session_state_from_auth_response = (known after apply) + frontchannel_logout_enabled = (known after apply) + full_scope_allowed = true + id = (known after apply) + implicit_flow_enabled = (known after apply) + import = false + name = "study-staging" + oauth2_device_authorization_grant_enabled = false + oauth2_jwt_authorization_grant_enabled = false + realm_id = "scottylabs" + require_dpop_bound_tokens = (known after apply) + resource_server_id = (known after apply) + service_account_user_id = (known after apply) + service_accounts_enabled = (known after apply) + standard_flow_enabled = true + standard_token_exchange_enabled = false + use_refresh_tokens = true + use_refresh_tokens_client_credentials = false + valid_post_logout_redirect_uris = (known after apply) + valid_redirect_uris = [ + "https://oauth.scottylabs.org/oauth2/callback", ] + web_origins = [ + "+", ] } # keycloak_openid_group_membership_protocol_mapper.study will be created + resource "keycloak_openid_group_membership_protocol_mapper" "study" { + add_to_access_token = true + add_to_id_token = true + add_to_token_introspection = false + add_to_userinfo = true + claim_name = "groups" + client_id = (known after apply) + full_path = true + id = (known after apply) + name = "groups" + realm_id = "scottylabs" } # keycloak_openid_group_membership_protocol_mapper.study_dev will be created + resource "keycloak_openid_group_membership_protocol_mapper" "study_dev" { + add_to_access_token = true + add_to_id_token = true + add_to_token_introspection = false + add_to_userinfo = true + claim_name = "groups" + client_id = (known after apply) + full_path = true + id = (known after apply) + name = "groups" + realm_id = "scottylabs" } # keycloak_openid_group_membership_protocol_mapper.study_staging will be created + resource "keycloak_openid_group_membership_protocol_mapper" "study_staging" { + add_to_access_token = true + add_to_id_token = true + add_to_token_introspection = false + add_to_userinfo = true + claim_name = "groups" + client_id = (known after apply) + full_path = true + id = (known after apply) + name = "groups" + realm_id = "scottylabs" } # vault_kv_secret_v2.study_dev_keycloak_realm will be created + resource "vault_kv_secret_v2" "study_dev_keycloak_realm" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/dev/KEYCLOAK_REALM" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_dev_keycloak_url will be created + resource "vault_kv_secret_v2" "study_dev_keycloak_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/dev/KEYCLOAK_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_dev_oauth_relay_url will be created + resource "vault_kv_secret_v2" "study_dev_oauth_relay_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/dev/OAUTH_RELAY_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_dev_oidc_client_id will be created + resource "vault_kv_secret_v2" "study_dev_oidc_client_id" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/dev/OIDC_CLIENT_ID" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_dev_oidc_client_secret will be created + resource "vault_kv_secret_v2" "study_dev_oidc_client_secret" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/dev/OIDC_CLIENT_SECRET" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_dev_project_admin_group will be created + resource "vault_kv_secret_v2" "study_dev_project_admin_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/dev/PROJECT_ADMIN_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_dev_project_group will be created + resource "vault_kv_secret_v2" "study_dev_project_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/dev/PROJECT_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_preview_keycloak_realm will be created + resource "vault_kv_secret_v2" "study_preview_keycloak_realm" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/preview/KEYCLOAK_REALM" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_preview_keycloak_url will be created + resource "vault_kv_secret_v2" "study_preview_keycloak_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/preview/KEYCLOAK_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_preview_oauth_relay_url will be created + resource "vault_kv_secret_v2" "study_preview_oauth_relay_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/preview/OAUTH_RELAY_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_preview_oidc_client_id will be created + resource "vault_kv_secret_v2" "study_preview_oidc_client_id" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/preview/OIDC_CLIENT_ID" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_preview_oidc_client_secret will be created + resource "vault_kv_secret_v2" "study_preview_oidc_client_secret" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/preview/OIDC_CLIENT_SECRET" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_preview_project_admin_group will be created + resource "vault_kv_secret_v2" "study_preview_project_admin_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/preview/PROJECT_ADMIN_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_preview_project_group will be created + resource "vault_kv_secret_v2" "study_preview_project_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/preview/PROJECT_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_prod_keycloak_realm will be created + resource "vault_kv_secret_v2" "study_prod_keycloak_realm" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/KEYCLOAK_REALM" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_prod_keycloak_url will be created + resource "vault_kv_secret_v2" "study_prod_keycloak_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/KEYCLOAK_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_prod_oauth_relay_url will be created + resource "vault_kv_secret_v2" "study_prod_oauth_relay_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/OAUTH_RELAY_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_prod_oidc_client_id will be created + resource "vault_kv_secret_v2" "study_prod_oidc_client_id" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/OIDC_CLIENT_ID" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_prod_oidc_client_secret will be created + resource "vault_kv_secret_v2" "study_prod_oidc_client_secret" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/OIDC_CLIENT_SECRET" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_prod_project_admin_group will be created + resource "vault_kv_secret_v2" "study_prod_project_admin_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/PROJECT_ADMIN_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_prod_project_group will be created + resource "vault_kv_secret_v2" "study_prod_project_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/prod/PROJECT_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_staging_keycloak_realm will be created + resource "vault_kv_secret_v2" "study_staging_keycloak_realm" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/staging/KEYCLOAK_REALM" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_staging_keycloak_url will be created + resource "vault_kv_secret_v2" "study_staging_keycloak_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/staging/KEYCLOAK_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_staging_oauth_relay_url will be created + resource "vault_kv_secret_v2" "study_staging_oauth_relay_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/staging/OAUTH_RELAY_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_staging_oidc_client_id will be created + resource "vault_kv_secret_v2" "study_staging_oidc_client_id" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/staging/OIDC_CLIENT_ID" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_staging_oidc_client_secret will be created + resource "vault_kv_secret_v2" "study_staging_oidc_client_secret" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/staging/OIDC_CLIENT_SECRET" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_staging_project_admin_group will be created + resource "vault_kv_secret_v2" "study_staging_project_admin_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/staging/PROJECT_ADMIN_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.study_staging_project_group will be created + resource "vault_kv_secret_v2" "study_staging_project_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/study/staging/PROJECT_GROUP" + path = (known after apply) + custom_metadata (known after apply) } Plan: 38 to add, 1 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 38 to add, 1 to change, 0 to destroy. --- ### 9. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.28s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_channel_permission.cmu-study_ch0 will be created + resource "discord_channel_permission" "cmu-study_ch0" { + allow = 1024 + channel_id = "1512270354943578132" + id = (known after apply) + overwrite_id = (known after apply) + type = "role" } # discord_member_roles.annadavi will be created + resource "discord_member_roles" "annadavi" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "991766392590311455" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1519792320755662928" } + role { + has_role = true + role_id = (known after apply) } } # discord_role.cmu-study will be created + resource "discord_role" "cmu-study" { + hoist = false + id = (known after apply) + managed = (known after apply) + mentionable = false + name = "CMU Study" + permissions = 0 + position = (known after apply) + server_id = "1353118209024331948" } Plan: 3 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 3 to add, 0 to change, 0 to destroy. --- ### 10. project: `git` dir: `tofu/git` workspace: `default` **Plan Error** <details><summary>Show Output</summary> ``` running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu plan -input=false -refresh -out /var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git/git-default.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git': exit status 1 Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ random_password.kennel_webhook_secret: Refreshing state... [id=none] github_repository.quest_quest: Refreshing state... [id=quest] github_team.cmu-housing: Refreshing state... [id=17937919] github_repository.slai_cmugpt_sms_surface: Refreshing state... [id=cmugpt-sms-surface] github_repository.cmu-insta_insta: Refreshing state... [id=insta] github_team.cmu-insta: Refreshing state... [id=19240311] github_repository.cmu-maps_maps: Refreshing state... [id=maps] github_repository.uia_scottylabs_org: Refreshing state... [id=scottylabs-org] github_repository.slai_cmugpt_agent: Refreshing state... [id=cmugpt-agent] github_repository.cbp_dalmatian: Refreshing state... [id=dalmatian] github_repository.lost-and-found_lost_and_found_v2: Refreshing state... [id=lost-and-found-v2] github_team.terrier: Refreshing state... [id=12028577] github_repository.cmu-housing_housing: Refreshing state... [id=housing] github_repository.cmu-cal_cal: Refreshing state... [id=cal] github_team.occ: Refreshing state... [id=19240310] github_team.devops: Refreshing state... [id=14706218] github_team.lost-and-found: Refreshing state... [id=4088608] github_team.cmu-maps: Refreshing state... [id=10872360] github_repository.devops_infrastructure: Refreshing state... [id=infrastructure] github_repository.occ_occ: Refreshing state... [id=occ] github_repository.devops_documentation: Refreshing state... [id=documentation] github_repository.cmu-courses_courses: Refreshing state... [id=courses] github_repository.slai_cmugpt_surface: Refreshing state... [id=cmugpt-surface] github_repository.cmu-courses_internet_archive: Refreshing state... [id=internet-archive] github_repository.cmu-eats_dining_api: Refreshing state... [id=dining-api] github_repository.terrier_terrier: Refreshing state... [id=terrier] github_team.cmu-courses: Refreshing state... [id=3643907] github_team.cmu-eats: Refreshing state... [id=16807600] github_repository.devops_link_shortener: Refreshing state... [id=link-shortener] github_repository.cbp_groupme_mirror: Refreshing state... [id=groupme-mirror] github_repository.uia_components: Refreshing state... [id=components] github_team.quest: Refreshing state... [id=17330197] github_repository.devops_kennel: Refreshing state... [id=kennel] github_repository.devops_collie: Refreshing state... [id=collie] github_repository.cbp_bus_sign: Refreshing state... [id=bus-sign] github_repository.cbp_discord_verify: Refreshing state... [id=discord-verify] github_team.cbp: Refreshing state... [id=17170099] github_team.tartan-vote: Refreshing state... [id=17701833] github_team.cmu-cal: Refreshing state... [id=15669112] github_team.uia: Refreshing state... [id=17529763] github_repository.devops_governance: Refreshing state... [id=governance] github_team.slai: Refreshing state... [id=17571328] github_repository.cmu-eats_eats: Refreshing state... [id=eats] github_repository.tartan-vote_tartan_vote: Refreshing state... [id=tartan-vote] github_repository.slai_mcp_server: Refreshing state... [id=mcp-server] forgejo_repository.devops_infrastructure: Refreshing state... [name=infrastructure] forgejo_repository.devops_collie: Refreshing state... [name=collie] forgejo_repository.quest_quest: Refreshing state... [name=quest] forgejo_repository.cmu-cal_cal: Refreshing state... [name=cal] forgejo_repository.tartan-vote_tartan_vote: Refreshing state... [name=tartan-vote] forgejo_repository.uia_scottylabs_org: Refreshing state... [name=scottylabs-org] forgejo_repository.cmu-courses_internet_archive: Refreshing state... [name=internet-archive] forgejo_repository.devops_kennel: Refreshing state... [name=kennel] forgejo_repository.devops_link_shortener: Refreshing state... [name=link-shortener] forgejo_repository.slai_mcp_server: Refreshing state... [name=mcp-server] forgejo_repository.cbp_dalmatian: Refreshing state... [name=dalmatian] forgejo_repository.lost-and-found_lost_and_found_v2: Refreshing state... [name=lost-and-found-v2] forgejo_repository.cmu-courses_courses: Refreshing state... [name=courses] forgejo_repository.cbp_groupme_mirror: Refreshing state... [name=groupme-mirror] forgejo_repository.cmu-eats_eats: Refreshing state... [name=eats] forgejo_repository.cbp_bus_sign: Refreshing state... [name=bus-sign] forgejo_repository.occ_occ: Refreshing state... [name=occ] data.forgejo_organization.this: Reading... data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs] forgejo_repository.slai_cmugpt_sms_surface: Refreshing state... [name=cmugpt-sms-surface] forgejo_repository.cmu-eats_dining_api: Refreshing state... [name=dining-api] forgejo_repository.devops_documentation: Refreshing state... [name=documentation] forgejo_repository.cmu-insta_insta: Refreshing state... [name=insta] forgejo_repository.slai_cmugpt_surface: Refreshing state... [name=cmugpt-surface] forgejo_repository.terrier_terrier: Refreshing state... [name=terrier] forgejo_repository.slai_cmugpt_agent: Refreshing state... [name=cmugpt-agent] forgejo_repository.cbp_discord_verify: Refreshing state... [name=discord-verify] forgejo_repository.devops_governance: Refreshing state... [name=governance] forgejo_repository.cmu-housing_housing: Refreshing state... [name=housing] forgejo_repository.uia_components: Refreshing state... [name=components] forgejo_repository.cmu-maps_maps: Refreshing state... [name=maps] data.external.identity_acchs: Reading... data.external.identity_nikhilr: Reading... data.external.identity_dcrodrig: Reading... data.external.identity_xboxbedrock: Reading... data.external.identity_acchs: Read complete after 0s [id=-] data.external.identity_brightz: Reading... data.external.identity_dcrodrig: Read complete after 0s [id=-] data.external.identity_qianxuege: Reading... data.external.identity_nikhilr: Read complete after 0s [id=-] data.external.identity_maybe_yiyi: Reading... data.external.identity_dkou: Reading... data.external.identity_xboxbedrock: Read complete after 1s [id=-] data.external.identity_jc_965: Reading... data.external.identity_qianxuege: Read complete after 1s [id=-] data.external.identity_gostmeaper: Reading... data.external.identity_maybe_yiyi: Read complete after 1s [id=-] data.external.identity_kothurvidur: Reading... data.external.identity_brightz: Read complete after 1s [id=-] data.external.identity_manrajs: Reading... data.external.identity_emmawong: Reading... data.external.identity_dkou: Read complete after 1s [id=-] data.external.identity_jalenluorion: Reading... data.external.identity_jc_965: Read complete after 0s [id=-] data.external.identity_thesuperrl: Reading... data.external.identity_kothurvidur: Read complete after 0s [id=-] data.external.identity_riztofu: Reading... data.external.identity_manrajs: Read complete after 0s [id=-] data.external.identity_arom1a: Reading... data.external.identity_gostmeaper: Read complete after 0s [id=-] data.external.identity_liamwalk: Reading... data.external.identity_emmawong: Read complete after 0s [id=-] data.external.identity_stringatlas: Reading... data.external.identity_jalenluorion: Read complete after 0s [id=-] data.external.identity_rzhu3: Reading... data.external.identity_thesuperrl: Read complete after 0s [id=-] data.external.identity_amzoeee: Reading... data.external.identity_arom1a: Read complete after 0s [id=-] data.external.identity_krishsax: Reading... data.external.identity_liamwalk: Read complete after 0s [id=-] data.external.identity_jw8: Reading... data.external.identity_riztofu: Read complete after 0s [id=-] data.external.identity_rhsen: Reading... data.external.identity_stringatlas: Read complete after 0s [id=-] data.external.identity_jessica_lr_song: Reading... data.external.identity_rzhu3: Read complete after 0s [id=-] data.external.identity_anish: Reading... data.external.identity_amzoeee: Read complete after 0s [id=-] data.external.identity_annadavi: Reading... data.external.identity_jw8: Read complete after 0s [id=-] data.external.identity_siddharthr: Reading... data.external.identity_krishsax: Read complete after 0s [id=-] data.external.identity_gabrielx: Reading... data.external.identity_rhsen: Read complete after 0s [id=-] data.external.identity_kluoxx: Reading... data.external.identity_jessica_lr_song: Read complete after 0s [id=-] data.external.identity_tentype: Reading... data.external.identity_anish: Read complete after 0s [id=-] data.external.identity_kritdass: Reading... data.external.identity_annadavi: Read complete after 0s [id=-] data.external.identity_jiyas: Reading... data.external.identity_kluoxx: Read complete after 1s [id=-] data.external.identity_zoexue: Reading... data.external.identity_siddharthr: Read complete after 1s [id=-] data.external.identity_kechezon: Reading... data.external.identity_gabrielx: Read complete after 1s [id=-] data.external.identity_etashjha: Reading... data.external.identity_tentype: Read complete after 1s [id=-] github_actions_repository_permissions.slai_cmugpt_sms_surface_actions: Refreshing state... [id=cmugpt-sms-surface] data.external.identity_kritdass: Read complete after 1s [id=-] github_actions_repository_permissions.slai_cmugpt_agent_actions: Refreshing state... [id=cmugpt-agent] data.external.identity_jiyas: Read complete after 1s [id=-] github_actions_repository_permissions.lost-and-found_lost_and_found_v2_actions: Refreshing state... [id=lost-and-found-v2] github_actions_repository_permissions.quest_quest_actions: Refreshing state... [id=quest] data.external.identity_zoexue: Read complete after 0s [id=-] github_actions_repository_permissions.cmu-maps_maps_actions: Refreshing state... [id=maps] data.external.identity_kechezon: Read complete after 0s [id=-] github_actions_repository_permissions.cmu-insta_insta_actions: Refreshing state... [id=insta] data.external.identity_etashjha: Read complete after 0s [id=-] github_actions_repository_permissions.uia_scottylabs_org_actions: Refreshing state... [id=scottylabs-org] github_actions_repository_permissions.cbp_dalmatian_actions: Refreshing state... [id=dalmatian] github_actions_repository_permissions.cmu-housing_housing_actions: Refreshing state... [id=housing] github_actions_repository_permissions.cmu-cal_cal_actions: Refreshing state... [id=cal] github_actions_repository_permissions.cmu-courses_courses_actions: Refreshing state... [id=courses] github_actions_repository_permissions.slai_cmugpt_surface_actions: Refreshing state... [id=cmugpt-surface] github_actions_repository_permissions.cmu-courses_internet_archive_actions: Refreshing state... [id=internet-archive] github_actions_repository_permissions.devops_infrastructure_actions: Refreshing state... [id=infrastructure] github_actions_repository_permissions.occ_occ_actions: Refreshing state... [id=occ] github_actions_repository_permissions.devops_documentation_actions: Refreshing state... [id=documentation] github_actions_repository_permissions.cbp_groupme_mirror_actions: Refreshing state... [id=groupme-mirror] github_actions_repository_permissions.uia_components_actions: Refreshing state... [id=components] github_actions_repository_permissions.cmu-eats_dining_api_actions: Refreshing state... [id=dining-api] github_actions_repository_permissions.terrier_terrier_actions: Refreshing state... [id=terrier] github_actions_repository_permissions.devops_collie_actions: Refreshing state... [id=collie] github_actions_repository_permissions.cbp_discord_verify_actions: Refreshing state... [id=discord-verify] github_actions_repository_permissions.devops_link_shortener_actions: Refreshing state... [id=link-shortener] restapi_object.devops_infrastructure_push_mirror: Refreshing state... [id=remote_mirror_oYtFpKVzTaZ] restapi_object.devops_infrastructure_kennel_webhook: Refreshing state... [id=26] restapi_object.devops_collie_push_mirror: Refreshing state... [id=remote_mirror_ljvVejNFXaS] restapi_object.devops_collie_kennel_webhook: Refreshing state... [id=17] restapi_object.quest_quest_kennel_webhook: Refreshing state... [id=10] restapi_object.quest_quest_push_mirror: Refreshing state... [id=remote_mirror_U9d9BPAon29] restapi_object.cmu-cal_cal_push_mirror: Refreshing state... [id=remote_mirror_CO243uIONoq] restapi_object.cmu-cal_cal_kennel_webhook: Refreshing state... [id=20] restapi_object.tartan-vote_tartan_vote_kennel_webhook: Refreshing state... [id=24] restapi_object.tartan-vote_tartan_vote_push_mirror: Refreshing state... [id=remote_mirror_EUFq4sFSnH5] restapi_object.tartan-vote_tartan_vote_docs_webhook: Refreshing state... [id=25] restapi_object.uia_scottylabs_org_kennel_webhook: Refreshing state... [id=29] restapi_object.uia_scottylabs_org_push_mirror: Refreshing state... [id=remote_mirror_h6yN6QdS1lD] restapi_object.cmu-courses_internet_archive_push_mirror: Refreshing state... [id=remote_mirror_58X__Yfs0bR] restapi_object.devops_link_shortener_kennel_webhook: Refreshing state... [id=8] restapi_object.devops_link_shortener_push_mirror: Refreshing state... [id=remote_mirror_cp8daMpt82v] restapi_object.devops_kennel_kennel_webhook: Refreshing state... [id=14] restapi_object.devops_kennel_push_mirror: Refreshing state... [id=remote_mirror_cT9psJkhTQ4] restapi_object.slai_mcp_server_push_mirror: Refreshing state... [id=remote_mirror_Tqh2y0R_vDp] restapi_object.slai_mcp_server_kennel_webhook: Refreshing state... [id=13] restapi_object.cbp_dalmatian_kennel_webhook: Refreshing state... [id=6] restapi_object.cbp_dalmatian_push_mirror: Refreshing state... [id=remote_mirror_qzlCpLK153F] restapi_object.cbp_dalmatian_docs_webhook: Refreshing state... [id=7] restapi_object.lost-and-found_lost_and_found_v2_push_mirror: Refreshing state... [id=remote_mirror_n_UZBFU-CY4] restapi_object.cmu-courses_courses_push_mirror: Refreshing state... [id=remote_mirror_WMR9yQwKFDF] restapi_object.cmu-courses_courses_kennel_webhook: Refreshing state... [id=11] restapi_object.cbp_groupme_mirror_push_mirror: Refreshing state... [id=remote_mirror_U61HJHB9VGE] restapi_object.cmu-eats_eats_push_mirror: Refreshing state... [id=remote_mirror_sjYwMA6NPgh] restapi_object.cmu-eats_eats_kennel_webhook: Refreshing state... [id=31] restapi_object.cbp_bus_sign_push_mirror: Refreshing state... [id=remote_mirror_MkKxQ2vA9pc] restapi_object.cbp_bus_sign_kennel_webhook: Refreshing state... [id=21] forgejo_team.devops: Refreshing state... [name=devops] forgejo_team.bus-sign_leads: Refreshing state... [name=bus-sign-leads] forgejo_team.courses_leads: Refreshing state... [name=courses-leads] forgejo_team.cmu-courses: Refreshing state... [name=cmu-courses] forgejo_team.scottylabs-org_leads: Refreshing state... [name=scottylabs-org-leads] forgejo_team.dalmatian_leads: Refreshing state... [name=dalmatian-leads] forgejo_team.cmu-cal: Refreshing state... [name=cmu-cal] forgejo_team.cmugpt-surface_leads: Refreshing state... [name=cmugpt-surface-leads] forgejo_team.tartan-vote_leads: Refreshing state... [name=tartan-vote-leads] forgejo_team.insta_leads: Refreshing state... [name=insta-leads] forgejo_team.link-shortener_leads: Refreshing state... [name=link-shortener-leads] forgejo_team.lost-and-found-v2_leads: Refreshing state... [name=lost-and-found-v2-leads] forgejo_team.discord-verify_leads: Refreshing state... [name=discord-verify-leads] forgejo_team.cmugpt-sms-surface_leads: Refreshing state... [name=cmugpt-sms-surface-leads] forgejo_team.quest: Refreshing state... [name=quest] forgejo_team.lost-and-found: Refreshing state... [name=lost-and-found] forgejo_team.tartan-vote: Refreshing state... [name=tartan-vote] forgejo_team.occ-main_leads: Refreshing state... [name=occ-main-leads] forgejo_team.documentation_leads: Refreshing state... [name=documentation-leads] forgejo_team.cal_leads: Refreshing state... [name=cal-leads] forgejo_team.cmu-housing: Refreshing state... [name=cmu-housing] forgejo_team.terrier: Refreshing state... [name=terrier] forgejo_team.cmugpt-mcp-server_leads: Refreshing state... [name=cmugpt-mcp-server-leads] forgejo_team.cmugpt-agent_leads: Refreshing state... [name=cmugpt-agent-leads] forgejo_team.groupme-mirror_leads: Refreshing state... [name=groupme-mirror-leads] forgejo_team.cmu-insta: Refreshing state... [name=cmu-insta] forgejo_team.cmu-courses_leads: Refreshing state... [name=cmu-courses-leads] forgejo_team.components_leads: Refreshing state... [name=components-leads] forgejo_team.cbp: Refreshing state... [name=cbp] forgejo_team.housing_leads: Refreshing state... [name=housing-leads] forgejo_team.devops_leads: Refreshing state... [name=devops-leads] forgejo_team.uia: Refreshing state... [name=uia] forgejo_team.terrier_leads: Refreshing state... [name=terrier-leads] forgejo_team.slai: Refreshing state... [name=slai] forgejo_team.maps_leads: Refreshing state... [name=maps-leads] forgejo_team.quest_leads: Refreshing state... [name=quest-leads] forgejo_team.cmu-maps: Refreshing state... [name=cmu-maps] forgejo_team.eats_leads: Refreshing state... [name=eats-leads] forgejo_team.occ: Refreshing state... [name=occ] forgejo_team.cmu-eats: Refreshing state... [name=cmu-eats] restapi_object.occ_occ_push_mirror: Refreshing state... [id=remote_mirror_moBLq1YG1qz] restapi_object.devops_documentation_push_mirror: Refreshing state... [id=remote_mirror_izmsqaNB6Wx] restapi_object.slai_cmugpt_sms_surface_push_mirror: Refreshing state... [id=remote_mirror__fkBkASN2jA] restapi_object.slai_cmugpt_sms_surface_kennel_webhook: Refreshing state... [id=19] restapi_object.cmu-eats_dining_api_push_mirror: Refreshing state... [id=remote_mirror_a4MsQ5rO1_Z] restapi_object.slai_cmugpt_surface_kennel_webhook: Refreshing state... [id=23] restapi_object.slai_cmugpt_surface_push_mirror: Refreshing state... [id=remote_mirror_cPTQN6S5SpB] restapi_object.cmu-insta_insta_push_mirror: Refreshing state... [id=remote_mirror_891nu6XQ1us] restapi_object.terrier_terrier_push_mirror: Refreshing state... [id=remote_mirror_YopDgZksZv5] restapi_object.terrier_terrier_kennel_webhook: Refreshing state... [id=12] restapi_object.slai_cmugpt_agent_push_mirror: Refreshing state... [id=remote_mirror_kNY6rVcxdyr] restapi_object.slai_cmugpt_agent_kennel_webhook: Refreshing state... [id=28] restapi_object.devops_governance_kennel_webhook: Refreshing state... [id=22] restapi_object.devops_governance_push_mirror: Refreshing state... [id=remote_mirror_lt9JzkX1jMq] restapi_object.cbp_discord_verify_push_mirror: Refreshing state... [id=remote_mirror_3MV7Up6cY1X] restapi_object.cbp_discord_verify_kennel_webhook: Refreshing state... [id=18] restapi_object.cmu-housing_housing_kennel_webhook: Refreshing state... [id=16] restapi_object.cmu-housing_housing_docs_webhook: Refreshing state... [id=15] restapi_object.cmu-housing_housing_push_mirror: Refreshing state... [id=remote_mirror_uoV2HG4aiOT] restapi_object.uia_components_push_mirror: Refreshing state... [id=remote_mirror_Kwwa8TDnTgc] restapi_object.uia_components_kennel_webhook: Refreshing state... [id=27] restapi_object.cmu-maps_maps_kennel_webhook: Refreshing state... [id=9] restapi_object.cmu-maps_maps_push_mirror: Refreshing state... [id=remote_mirror_e4_FUnvoslJ] github_actions_repository_permissions.devops_kennel_actions: Refreshing state... [id=kennel] github_team_membership.occ_acchs: Refreshing state... [id=19240310:acchs-at-cmu] github_team_membership.occ_dcrodrig: Refreshing state... [id=19240310:newhacker1746] github_team_membership.cmu-housing_nikhilr: Refreshing state... [id=17937919:ecstaticpilot] github_team_membership.cbp_nikhilr: Refreshing state... [id=17170099:ecstaticpilot] github_team_membership.quest_xboxbedrock: Refreshing state... [id=17330197:xboxbedrock] github_team_membership.cbp_xboxbedrock: Refreshing state... [id=17170099:xboxbedrock] github_team_membership.tartan-vote_xboxbedrock: Refreshing state... [id=17701833:xboxbedrock] github_team_membership.cmu-housing_xboxbedrock: Refreshing state... [id=17937919:xboxbedrock] github_team_membership.devops_xboxbedrock: Refreshing state... [id=14706218:xboxbedrock] github_team_membership.occ_xboxbedrock: Refreshing state... [id=19240310:xboxbedrock] github_team_membership.slai_xboxbedrock: Refreshing state... [id=17571328:xboxbedrock] github_team_membership.slai_qianxuege: Refreshing state... [id=17571328:qianxuege] github_team_membership.cbp_maybe_yiyi: Refreshing state... [id=17170099:maybe-yiyi] github_team_membership.tartan-vote_maybe_yiyi: Refreshing state... [id=17701833:maybe-yiyi] github_team_membership.quest_brightz: Refreshing state... [id=17330197:bright-zheng] github_actions_repository_permissions.cbp_bus_sign_actions: Refreshing state... [id=bus-sign] github_team_membership.cmu-courses_dkou: Refreshing state... [id=3643907:sharkree] github_team_membership.terrier_dkou: Refreshing state... [id=12028577:sharkree] github_team_membership.slai_jc_965: Refreshing state... [id=17571328:jc-965] github_team_membership.terrier_kothurvidur: Refreshing state... [id=12028577:kothurvidur] github_team_membership.cmu-housing_manrajs: Refreshing state... [id=17937919:manraj10] github_team_membership.cmu-housing_gostmeaper: Refreshing state... [id=17937919:gostmeaper] github_team_membership.cmu-housing_emmawong: Refreshing state... [id=17937919:emw868] github_team_membership.cmu-cal_jalenluorion: Refreshing state... [id=15669112:jalenluorion] github_team_membership.devops_thesuperrl: Refreshing state... [id=14706218:thesuperrl] github_team_membership.cmu-eats_arom1a: Refreshing state... [id=16807600:arom1a] github_team_membership.cbp_liamwalk: Refreshing state... [id=17170099:pandaaured] github_team_membership.tartan-vote_riztofu: Refreshing state... [id=17701833:riztofu] github_team_membership.cmu-housing_stringatlas: Refreshing state... [id=17937919:stringatlas] github_team_membership.slai_rzhu3: Refreshing state... [id=17571328:richard-31415] github_team_membership.terrier_amzoeee: Refreshing state... [id=12028577:amzoeee] github_team_membership.cmu-housing_amzoeee: Refreshing state... [id=17937919:amzoeee] github_team_membership.quest_jw8: Refreshing state... [id=17330197:jeffreyyijiwang] github_team_membership.slai_krishsax: Refreshing state... [id=17571328:krishsaxena] github_team_membership.cmu-housing_rhsen: Refreshing state... [id=17937919:rohansen08] github_team_membership.cmu-housing_jessica_lr_song: Refreshing state... [id=17937919:jessica-lr-song] github_team_membership.slai_anish: Refreshing state... [id=17571328:ap-1] github_team_membership.terrier_anish: Refreshing state... [id=12028577:ap-1] github_team_membership.cmu-housing_anish: Refreshing state... [id=17937919:ap-1] github_team_membership.cmu-cal_anish: Refreshing state... [id=15669112:ap-1] github_team_membership.occ_anish: Refreshing state... [id=19240310:ap-1] github_team_membership.tartan-vote_anish: Refreshing state... [id=17701833:ap-1] github_team_membership.quest_anish: Refreshing state... [id=17330197:ap-1] github_team_membership.cmu-courses_anish: Refreshing state... [id=3643907:ap-1] github_team_membership.cmu-maps_anish: Refreshing state... [id=10872360:ap-1] github_team_membership.lost-and-found_anish: Refreshing state... [id=4088608:ap-1] github_team_membership.devops_anish: Refreshing state... [id=14706218:ap-1] github_team_membership.cbp_anish: Refreshing state... [id=17170099:ap-1] github_team_membership.uia_anish: Refreshing state... [id=17529763:ap-1] github_team_membership.cmu-housing_kluoxx: Refreshing state... [id=17937919:luokaden888] github_team_membership.cmu-maps_siddharthr: Refreshing state... [id=10872360:overthrowing] github_team_membership.cmu-courses_gabrielx: Refreshing state... [id=3643907:gabriel-xu] github_team_membership.cmu-housing_tentype: Refreshing state... [id=17937919:tentype] github_team_membership.cbp_tentype: Refreshing state... [id=17170099:tentype] github_team_membership.tartan-vote_tentype: Refreshing state... [id=17701833:tentype] github_team_membership.terrier_kritdass: Refreshing state... [id=12028577:kritdass] github_team_membership.slai_jiyas: Refreshing state... [id=17571328:jiya2301] github_actions_repository_permissions.devops_governance_actions: Refreshing state... [id=governance] github_team_membership.cmu-housing_zoexue: Refreshing state... [id=17937919:o-zox-o] github_team_membership.quest_kechezon: Refreshing state... [id=17330197:kechezon] github_team_membership.terrier_etashjha: Refreshing state... [id=12028577:etashj] github_team_membership.cmu-courses_etashjha: Refreshing state... [id=3643907:etashj] github_team_membership.cmu-insta_etashjha: Refreshing state... [id=19240311:etashj] github_actions_repository_permissions.cmu-eats_eats_actions: Refreshing state... [id=eats] github_actions_repository_permissions.tartan-vote_tartan_vote_actions: Refreshing state... [id=tartan-vote] github_actions_repository_permissions.slai_mcp_server_actions: Refreshing state... [id=mcp-server] github_repository_deploy_key.devops_infrastructure_mirror_deploy_key: Refreshing state... [id=infrastructure:159665343] github_repository_deploy_key.devops_collie_mirror_deploy_key: Refreshing state... [id=collie:159665332] github_repository_deploy_key.quest_quest_mirror_deploy_key: Refreshing state... [id=quest:159665323] github_repository_deploy_key.cmu-cal_cal_mirror_deploy_key: Refreshing state... [id=cal:159665344] github_repository_deploy_key.tartan-vote_tartan_vote_mirror_deploy_key: Refreshing state... [id=tartan-vote:159665350] github_repository_deploy_key.uia_scottylabs_org_mirror_deploy_key: Refreshing state... [id=scottylabs-org:159665361] github_repository_deploy_key.cmu-courses_internet_archive_mirror_deploy_key: Refreshing state... [id=internet-archive:159665356] github_repository_deploy_key.devops_link_shortener_mirror_deploy_key: Refreshing state... [id=link-shortener:159665345] github_repository_deploy_key.devops_kennel_mirror_deploy_key: Refreshing state... [id=kennel:159665328] github_repository_deploy_key.slai_mcp_server_mirror_deploy_key: Refreshing state... [id=mcp-server:159665358] github_repository_deploy_key.cbp_dalmatian_mirror_deploy_key: Refreshing state... [id=dalmatian:159665367] github_repository_deploy_key.lost-and-found_lost_and_found_v2_mirror_deploy_key: Refreshing state... [id=lost-and-found-v2:159665337] github_repository_deploy_key.cmu-courses_courses_mirror_deploy_key: Refreshing state... [id=courses:159665327] github_repository_deploy_key.cbp_groupme_mirror_mirror_deploy_key: Refreshing state... [id=groupme-mirror:159665366] github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Refreshing state... [id=eats:161775995] github_repository_deploy_key.cbp_bus_sign_mirror_deploy_key: Refreshing state... [id=bus-sign:159665342] forgejo_team_repository.devops_link_shortener: Refreshing state... forgejo_team_member.devops_thesuperrl: Refreshing state... forgejo_team_repository.devops_kennel: Refreshing state... forgejo_team_member.devops_anish: Refreshing state... forgejo_team_member.devops_xboxbedrock: Refreshing state... forgejo_team_repository.devops_governance: Refreshing state... forgejo_team_repository.devops_infrastructure: Refreshing state... forgejo_team_repository.devops_collie: Refreshing state... forgejo_team_repository.devops_documentation: Refreshing state... forgejo_team_member.bus-sign_leads_maybe_yiyi: Refreshing state... forgejo_team_member.cmu-courses_dkou: Refreshing state... forgejo_team_member.cmu-courses_gabrielx: Refreshing state... forgejo_team_member.bus-sign_leads_anish: Refreshing state... forgejo_team_member.bus-sign_leads_tentype: Refreshing state... forgejo_team_member.cmu-courses_etashjha: Refreshing state... forgejo_team_member.cmu-courses_anish: Refreshing state... forgejo_team_repository.cmu-courses_internet_archive: Refreshing state... forgejo_team_repository.cmu-courses_courses: Refreshing state... forgejo_team_member.courses_leads_anish: Refreshing state... forgejo_team_member.courses_leads_etashjha: Refreshing state... forgejo_team_member.scottylabs-org_leads_anish: Refreshing state... forgejo_team_member.dalmatian_leads_tentype: Refreshing state... forgejo_team_member.dalmatian_leads_anish: Refreshing state... forgejo_team_member.dalmatian_leads_maybe_yiyi: Refreshing state... forgejo_team_member.cmu-cal_anish: Refreshing state... forgejo_team_repository.cmu-cal_cal: Refreshing state... forgejo_team_member.cmu-cal_jalenluorion: Refreshing state... forgejo_team_member.cmugpt-surface_leads_anish: Refreshing state... forgejo_team_member.cmugpt-surface_leads_krishsax: Refreshing state... forgejo_team_member.tartan-vote_leads_maybe_yiyi: Refreshing state... forgejo_team_member.tartan-vote_leads_anish: Refreshing state... forgejo_team_member.insta_leads_etashjha: Refreshing state... forgejo_team_member.lost-and-found-v2_leads_anish: Refreshing state... forgejo_team_member.link-shortener_leads_anish: Refreshing state... forgejo_team_member.link-shortener_leads_thesuperrl: Refreshing state... forgejo_team_member.cmugpt-sms-surface_leads_krishsax: Refreshing state... forgejo_team_member.cmugpt-sms-surface_leads_anish: Refreshing state... forgejo_team_member.discord-verify_leads_tentype: Refreshing state... forgejo_team_member.discord-verify_leads_anish: Refreshing state... forgejo_team_member.discord-verify_leads_maybe_yiyi: Refreshing state... forgejo_team_repository.lost-and-found_lost_and_found_v2: Refreshing state... forgejo_team_member.lost-and-found_anish: Refreshing state... forgejo_team_member.quest_jw8: Refreshing state... forgejo_team_repository.quest_quest: Refreshing state... forgejo_team_member.quest_anish: Refreshing state... forgejo_team_member.quest_kechezon: Refreshing state... forgejo_team_member.quest_brightz: Refreshing state... forgejo_team_member.quest_xboxbedrock: Refreshing state... forgejo_team_member.occ-main_leads_dcrodrig: Refreshing state... forgejo_team_member.tartan-vote_riztofu: Refreshing state... forgejo_team_member.occ-main_leads_anish: Refreshing state... forgejo_team_member.tartan-vote_anish: Refreshing state... forgejo_team_repository.tartan-vote_tartan_vote: Refreshing state... forgejo_team_member.tartan-vote_tentype: Refreshing state... forgejo_team_member.tartan-vote_maybe_yiyi: Refreshing state... forgejo_team_member.tartan-vote_xboxbedrock: Refreshing state... forgejo_team_member.documentation_leads_thesuperrl: Refreshing state... forgejo_team_member.cal_leads_anish: Refreshing state... forgejo_team_member.documentation_leads_anish: Refreshing state... forgejo_team_member.cal_leads_jalenluorion: Refreshing state... forgejo_team_repository.terrier_terrier: Refreshing state... forgejo_team_member.cmu-housing_gostmeaper: Refreshing state... forgejo_team_member.terrier_kritdass: Refreshing state... forgejo_team_member.cmu-housing_jessica_lr_song: Refreshing state... forgejo_team_member.cmu-housing_manrajs: Refreshing state... forgejo_team_member.cmu-housing_anish: Refreshing state... forgejo_team_member.cmu-housing_nikhilr: Refreshing state... forgejo_team_repository.cmu-housing_housing: Refreshing state... forgejo_team_member.cmu-housing_emmawong: Refreshing state... forgejo_team_member.cmu-housing_xboxbedrock: Refreshing state... forgejo_team_member.cmu-housing_rhsen: Refreshing state... forgejo_team_member.cmu-housing_stringatlas: Refreshing state... forgejo_team_member.cmu-housing_amzoeee: Refreshing state... forgejo_team_member.terrier_etashjha: Refreshing state... forgejo_team_member.terrier_anish: Refreshing state... forgejo_team_member.terrier_kothurvidur: Refreshing state... forgejo_team_member.terrier_amzoeee: Refreshing state... forgejo_team_member.terrier_dkou: Refreshing state... forgejo_team_member.cmu-housing_kluoxx: Refreshing state... forgejo_team_member.cmu-housing_tentype: Refreshing state... forgejo_team_member.cmugpt-mcp-server_leads_anish: Refreshing state... forgejo_team_member.cmugpt-mcp-server_leads_krishsax: Refreshing state... forgejo_team_member.cmu-housing_zoexue: Refreshing state... forgejo_team_member.cmugpt-agent_leads_anish: Refreshing state... forgejo_team_member.cmugpt-agent_leads_krishsax: Refreshing state... forgejo_team_member.groupme-mirror_leads_anish: Refreshing state... forgejo_team_member.groupme-mirror_leads_tentype: Refreshing state... forgejo_team_member.groupme-mirror_leads_maybe_yiyi: Refreshing state... forgejo_team_member.cmu-courses_leads_etashjha: Refreshing state... forgejo_team_member.cmu-courses_leads_anish: Refreshing state... forgejo_team_repository.cmu-insta_insta: Refreshing state... forgejo_team_member.cmu-insta_etashjha: Refreshing state... forgejo_team_member.components_leads_anish: Refreshing state... forgejo_team_member.cbp_liamwalk: Refreshing state... forgejo_team_member.cbp_xboxbedrock: Refreshing state... forgejo_team_member.cbp_nikhilr: Refreshing state... forgejo_team_repository.cbp_bus_sign: Refreshing state... forgejo_team_repository.cbp_discord_verify: Refreshing state... forgejo_team_repository.cbp_groupme_mirror: Refreshing state... forgejo_team_repository.cbp_dalmatian: Refreshing state... forgejo_team_member.cbp_maybe_yiyi: Refreshing state... forgejo_team_member.cbp_tentype: Refreshing state... forgejo_team_member.cbp_anish: Refreshing state... forgejo_team_member.housing_leads_tentype: Refreshing state... forgejo_team_member.housing_leads_nikhilr: Refreshing state... forgejo_team_member.housing_leads_anish: Refreshing state... forgejo_team_member.housing_leads_emmawong: Refreshing state... forgejo_team_member.housing_leads_gostmeaper: Refreshing state... forgejo_team_member.devops_leads_thesuperrl: Refreshing state... forgejo_team_member.devops_leads_anish: Refreshing state... forgejo_team_member.uia_anish: Refreshing state... forgejo_team_repository.uia_components: Refreshing state... forgejo_team_repository.uia_scottylabs_org: Refreshing state... forgejo_team_member.terrier_leads_kritdass: Refreshing state... forgejo_team_member.terrier_leads_anish: Refreshing state... forgejo_team_member.slai_xboxbedrock: Refreshing state... forgejo_team_member.slai_qianxuege: Refreshing state... forgejo_team_repository.slai_mcp_server: Refreshing state... forgejo_team_member.slai_krishsax: Refreshing state... forgejo_team_member.slai_jc_965: Refreshing state... forgejo_team_member.slai_jiyas: Refreshing state... forgejo_team_member.slai_rzhu3: Refreshing state... forgejo_team_repository.slai_cmugpt_agent: Refreshing state... forgejo_team_repository.slai_cmugpt_surface: Refreshing state... forgejo_team_repository.slai_cmugpt_sms_surface: Refreshing state... forgejo_team_member.slai_anish: Refreshing state... forgejo_team_member.quest_leads_anish: Refreshing state... forgejo_team_member.quest_leads_xboxbedrock: Refreshing state... forgejo_team_member.maps_leads_anish: Refreshing state... forgejo_team_member.maps_leads_siddharthr: Refreshing state... forgejo_team_repository.cmu-maps_maps: Refreshing state... forgejo_team_member.cmu-maps_siddharthr: Refreshing state... forgejo_team_member.cmu-maps_anish: Refreshing state... forgejo_team_member.eats_leads_arom1a: Refreshing state... forgejo_team_member.occ_acchs: Refreshing state... forgejo_team_member.occ_xboxbedrock: Refreshing state... forgejo_team_repository.occ_occ: Refreshing state... forgejo_team_member.occ_anish: Refreshing state... forgejo_team_member.occ_dcrodrig: Refreshing state... forgejo_team_repository.cmu-eats_eats: Refreshing state... forgejo_team_member.cmu-eats_arom1a: Refreshing state... github_repository_deploy_key.occ_occ_mirror_deploy_key: Refreshing state... [id=occ:161720242] github_repository_deploy_key.slai_cmugpt_sms_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-sms-surface:159665365] forgejo_team_repository.cmu-eats_dining_api: Refreshing state... github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Refreshing state... [id=dining-api:161772815] github_repository_deploy_key.devops_documentation_mirror_deploy_key: Refreshing state... [id=documentation:159665362] github_repository_deploy_key.slai_cmugpt_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-surface:159665348] github_repository_deploy_key.terrier_terrier_mirror_deploy_key: Refreshing state... [id=terrier:159665339] github_repository_deploy_key.cmu-insta_insta_mirror_deploy_key: Refreshing state... [id=insta:161720245] github_repository_deploy_key.cbp_discord_verify_mirror_deploy_key: Refreshing state... [id=discord-verify:159665340] github_repository_deploy_key.slai_cmugpt_agent_mirror_deploy_key: Refreshing state... [id=cmugpt-agent:159665360] github_repository_deploy_key.devops_governance_mirror_deploy_key: Refreshing state... [id=governance:159665346] github_repository_deploy_key.cmu-housing_housing_mirror_deploy_key: Refreshing state... [id=housing:159665369] github_repository_deploy_key.uia_components_mirror_deploy_key: Refreshing state... [id=components:159665353] github_repository_deploy_key.cmu-maps_maps_mirror_deploy_key: Refreshing state... [id=maps:159665325] github_repository_ruleset.devops_infrastructure_ruleset: Refreshing state... [id=11275118] github_repository_ruleset.devops_collie_ruleset: Refreshing state... [id=18865204] github_repository_ruleset.quest_quest_ruleset: Refreshing state... [id=15750052] github_repository_ruleset.cmu-cal_cal_ruleset: Refreshing state... [id=18919281] github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Refreshing state... [id=16848966] github_repository_ruleset.uia_scottylabs_org_ruleset: Refreshing state... [id=18112214] github_repository_ruleset.cmu-courses_internet_archive_ruleset: Refreshing state... [id=11988074] forgejo_team_repository.link-shortener_leads_link_shortener: Refreshing state... forgejo_team_repository.devops_leads_kennel: Refreshing state... forgejo_team_repository.devops_leads_governance: Refreshing state... forgejo_team_repository.devops_leads_infrastructure: Refreshing state... forgejo_team_repository.devops_leads_collie: Refreshing state... forgejo_team_repository.documentation_leads_documentation: Refreshing state... github_repository_ruleset.devops_link_shortener_ruleset: Refreshing state... [id=17911647] forgejo_team_repository.cmu-courses_leads_internet_archive: Refreshing state... forgejo_team_repository.courses_leads_courses: Refreshing state... forgejo_team_repository.cal_leads_cal: Refreshing state... github_repository_ruleset.devops_kennel_ruleset: Refreshing state... [id=13193453] forgejo_team_repository.lost-and-found-v2_leads_lost_and_found_v2: Refreshing state... forgejo_team_repository.quest_leads_quest: Refreshing state... forgejo_team_repository.tartan-vote_leads_tartan_vote: Refreshing state... github_repository_ruleset.slai_mcp_server_ruleset: Refreshing state... [id=17082154] forgejo_team_repository.terrier_leads_terrier: Refreshing state... forgejo_team_repository.housing_leads_housing: Refreshing state... github_repository_ruleset.cbp_dalmatian_ruleset: Refreshing state... [id=17238915] forgejo_team_repository.insta_leads_insta: Refreshing state... forgejo_team_repository.discord-verify_leads_discord_verify: Refreshing state... forgejo_team_repository.groupme-mirror_leads_groupme_mirror: Refreshing state... forgejo_team_repository.bus-sign_leads_bus_sign: Refreshing state... forgejo_team_repository.dalmatian_leads_dalmatian: Refreshing state... github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Refreshing state... [id=18920618] forgejo_team_repository.components_leads_components: Refreshing state... forgejo_team_repository.scottylabs-org_leads_scottylabs_org: Refreshing state... forgejo_team_repository.cmugpt-mcp-server_leads_mcp_server: Refreshing state... forgejo_team_repository.cmugpt-surface_leads_cmugpt_surface: Refreshing state... forgejo_team_repository.cmugpt-agent_leads_cmugpt_agent: Refreshing state... forgejo_team_repository.cmugpt-sms-surface_leads_cmugpt_sms_surface: Refreshing state... github_repository_ruleset.cmu-courses_courses_ruleset: Refreshing state... [id=15855161] forgejo_team_repository.maps_leads_maps: Refreshing state... forgejo_team_repository.occ-main_leads_occ: Refreshing state... github_repository_ruleset.cbp_groupme_mirror_ruleset: Refreshing state... [id=12503296] github_repository_ruleset.cmu-eats_eats_ruleset: Refreshing state... [id=21330218] github_repository_ruleset.cbp_bus_sign_ruleset: Refreshing state... [id=15236550] github_repository_ruleset.occ_occ_ruleset: Refreshing state... [id=21841634] github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Refreshing state... [id=17653622] github_repository_ruleset.cmu-eats_dining_api_ruleset: Refreshing state... [id=21879626] github_repository_ruleset.devops_documentation_ruleset: Refreshing state... [id=17079368] github_repository_ruleset.terrier_terrier_ruleset: Refreshing state... [id=12671003] github_repository_ruleset.slai_cmugpt_surface_ruleset: Refreshing state... [id=16476662] github_repository_ruleset.cmu-insta_insta_ruleset: Refreshing state... [id=21841636] github_repository_ruleset.cbp_discord_verify_ruleset: Refreshing state... [id=11449154] github_repository_ruleset.slai_cmugpt_agent_ruleset: Refreshing state... [id=16476661] github_repository_ruleset.devops_governance_ruleset: Refreshing state... [id=14488792] github_repository_ruleset.cmu-housing_housing_ruleset: Refreshing state... [id=17421604] github_repository_ruleset.uia_components_ruleset: Refreshing state... [id=16366085] github_repository_ruleset.cmu-maps_maps_ruleset: Refreshing state... [id=18864961] Note: Objects have changed outside of OpenTofu OpenTofu detected the following changes made outside of OpenTofu since the last "tofu apply" which may have affected this plan: # restapi_object.cmu-eats_dining_api_push_mirror has been deleted - resource "restapi_object" "cmu-eats_dining_api_push_mirror" { - api_data = { - "public_key" = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIIKyjdH3wH8J3aSgeZMLB9/uUGzyYltQsk5Cl1Q8MXqF" # (9 unchanged elements hidden) } -> null id = "remote_mirror_a4MsQ5rO1_Z" # (5 unchanged attributes hidden) } # restapi_object.cmu-eats_eats_push_mirror has been deleted - resource "restapi_object" "cmu-eats_eats_push_mirror" { - api_data = { - "public_key" = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFsm79Zlyi0JKH4y/BdgCCoLyrC4/18yF3owfsqOt53/" # (9 unchanged elements hidden) } -> null id = "remote_mirror_sjYwMA6NPgh" # (5 unchanged attributes hidden) } Unless you have made equivalent changes to your configuration, or ignored the relevant attributes using ignore_changes, the following plan may include actions to undo or respond to these changes. ───────────────────────────────────────────────────────────────────────────── OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ~ update in-place (current -> planned) -/+ destroy and then create replacement OpenTofu planned the following actions, but then encountered a problem: # forgejo_repository.cmu-eats_dining_api will be updated in-place ~ resource "forgejo_repository" "cmu-eats_dining_api" { ~ clone_url = "https://git.cmu.dev/ScottyLabs/dining-api.git" -> (known after apply) ~ created_at = "2026-09-18T23:16:27Z" -> (known after apply) + description = "Dining API for CMU dining locations and availability" ~ empty = false -> (known after apply) + external_tracker = (known after apply) + external_wiki = (known after apply) ~ forks_count = 0 -> (known after apply) ~ full_name = "ScottyLabs/dining-api" -> (known after apply) ~ html_url = "https://git.cmu.dev/ScottyLabs/dining-api" -> (known after apply) id = 97 ~ internal_tracker = { ~ allow_only_contributors_to_track_time = true -> (known after apply) ~ enable_issue_dependencies = true -> (known after apply) ~ enable_time_tracker = true -> (known after apply) } -> (known after apply) + mirror_interval = (known after apply) ~ mirror_updated = "0001-01-01T00:00:00Z" -> (known after apply) name = "dining-api" ~ open_issues_count = 0 -> (known after apply) ~ open_pr_counter = 5 -> (known after apply) ~ permissions = { ~ admin = true -> (known after apply) ~ pull = true -> (known after apply) ~ push = true -> (known after apply) } -> (known after apply) ~ release_counter = 0 -> (known after apply) ~ size = 38816 -> (known after apply) ~ ssh_url = "ssh://forgejo@git.cmu.dev/ScottyLabs/dining-api.git" -> (known after apply) ~ stars_count = 0 -> (known after apply) ~ updated_at = "2026-09-18T23:22:22Z" -> (known after apply) ~ watchers_count = 3 -> (known after apply) ~ website = "https://git.cmu.dev/ScottyLabs/dining-api" -> "https://api.cmueats.com" # (30 unchanged attributes hidden) } # forgejo_repository.cmu-eats_eats will be updated in-place ~ resource "forgejo_repository" "cmu-eats_eats" { ~ clone_url = "https://git.cmu.dev/ScottyLabs/eats.git" -> (known after apply) ~ created_at = "2026-09-18T23:15:39Z" -> (known after apply) ~ description = "Vite web app for CMU dining location info" -> "CMU dining location info" ~ empty = false -> (known after apply) + external_tracker = (known after apply) + external_wiki = (known after apply) ~ forks_count = 0 -> (known after apply) ~ full_name = "ScottyLabs/eats" -> (known after apply) ~ html_url = "https://git.cmu.dev/ScottyLabs/eats" -> (known after apply) id = 96 ~ internal_tracker = { ~ allow_only_contributors_to_track_time = true -> (known after apply) ~ enable_issue_dependencies = true -> (known after apply) ~ enable_time_tracker = true -> (known after apply) } -> (known after apply) + mirror_interval = (known after apply) ~ mirror_updated = "0001-01-01T00:00:00Z" -> (known after apply) name = "eats" ~ open_issues_count = 48 -> (known after apply) ~ open_pr_counter = 4 -> (known after apply) ~ permissions = { ~ admin = true -> (known after apply) ~ pull = true -> (known after apply) ~ push = true -> (known after apply) } -> (known after apply) ~ release_counter = 0 -> (known after apply) ~ size = 95496 -> (known after apply) ~ ssh_url = "ssh://forgejo@git.cmu.dev/ScottyLabs/eats.git" -> (known after apply) ~ stars_count = 0 -> (known after apply) ~ updated_at = "2026-09-18T23:35:12Z" -> (known after apply) ~ watchers_count = 3 -> (known after apply) # (31 unchanged attributes hidden) } # forgejo_repository.cmu-study_study will be created + resource "forgejo_repository" "cmu-study_study" { + allow_manual_merge = false + allow_merge_commits = true + allow_rebase = true + allow_rebase_explicit = true + allow_squash_merge = true + archive_on_destroy = false + archived = false + auto_init = true + autodetect_manual_merge = false + avatar_url = (known after apply) + clone_url = (known after apply) + created_at = (known after apply) + default_branch = "main" + default_merge_style = "merge" + description = "A platform for creating and joining study groups" + empty = (known after apply) + external_tracker = (known after apply) + external_wiki = (known after apply) + fork = (known after apply) + forks_count = (known after apply) + full_name = (known after apply) + has_actions = true + has_issues = true + has_packages = true + has_projects = true + has_pull_requests = true + has_releases = true + has_wiki = true + html_url = (known after apply) + id = (known after apply) + ignore_whitespace_conflicts = false + internal = (known after apply) + internal_tracker = (known after apply) + labels = false + lfs = false + milestones = false + mirror = false + mirror_interval = (known after apply) + mirror_updated = (known after apply) + name = "study" + open_issues_count = (known after apply) + open_pr_counter = (known after apply) + owner = "ScottyLabs" + parent_id = (known after apply) + permissions = (known after apply) + private = false + release_counter = (known after apply) + size = (known after apply) + ssh_url = (known after apply) + stars_count = (known after apply) + template = false + trust_model = "default" + updated_at = (known after apply) + watchers_count = (known after apply) + website = "https://cmustudy.com" } # forgejo_team.cmu-study will be created + resource "forgejo_team" "cmu-study" { + can_create_org_repo = (known after apply) + description = "CMU Study" + id = (known after apply) + includes_all_repositories = (known after apply) + name = "cmu-study" + organization = (known after apply) + organization_id = 3 + permission = "write" + units_map = { + "repo.code" = "write" + "repo.issues" = "write" + "repo.pulls" = "write" } } # forgejo_team.study_leads will be created + resource "forgejo_team" "study_leads" { + can_create_org_repo = false + description = "CMU Study leads" + id = (known after apply) + includes_all_repositories = false + name = "study-leads" + organization = (known after apply) + organization_id = 3 + permission = "admin" + units_map = { + "repo.actions" = "admin" + "repo.code" = "admin" + "repo.ext_issues" = "admin" + "repo.ext_wiki" = "admin" + "repo.issues" = "admin" + "repo.packages" = "admin" + "repo.projects" = "admin" + "repo.pulls" = "admin" + "repo.releases" = "admin" + "repo.wiki" = "admin" } } # forgejo_team_member.cmu-study_annadavi will be created + resource "forgejo_team_member" "cmu-study_annadavi" { + team_id = (known after apply) + user = "annadavi" } # forgejo_team_member.study_leads_annadavi will be created + resource "forgejo_team_member" "study_leads_annadavi" { + team_id = (known after apply) + user = "annadavi" } # forgejo_team_repository.cmu-study_study will be created + resource "forgejo_team_repository" "cmu-study_study" { + owner = "ScottyLabs" + repository = "study" + team_id = (known after apply) } # forgejo_team_repository.study_leads_study will be created + resource "forgejo_team_repository" "study_leads_study" { + owner = "ScottyLabs" + repository = "study" + team_id = (known after apply) } # github_actions_repository_permissions.cmu-study_study_actions will be created + resource "github_actions_repository_permissions" "cmu-study_study_actions" { + enabled = false + id = (known after apply) + repository = "study" + sha_pinning_required = (known after apply) } # github_repository.cmu-study_study will be created + resource "github_repository" "cmu-study_study" { + allow_auto_merge = false + allow_forking = (known after apply) + allow_merge_commit = true + allow_rebase_merge = true + allow_squash_merge = true + archived = false + default_branch = (known after apply) + delete_branch_on_merge = false + description = "READ-ONLY MIRROR" + etag = (known after apply) + fork = (known after apply) + full_name = (known after apply) + git_clone_url = (known after apply) + has_discussions = false + has_issues = false + has_projects = false + has_wiki = false + homepage_url = "https://git.cmu.dev/ScottyLabs/study" + html_url = (known after apply) + http_clone_url = (known after apply) + id = (known after apply) + ignore_vulnerability_alerts_during_read = false + merge_commit_message = "PR_TITLE" + merge_commit_title = "MERGE_MESSAGE" + name = "study" + node_id = (known after apply) + primary_language = (known after apply) + private = (known after apply) + repo_id = (known after apply) + source_owner = (known after apply) + source_repo = (known after apply) + squash_merge_commit_message = "COMMIT_MESSAGES" + squash_merge_commit_title = "COMMIT_OR_PR_TITLE" + ssh_clone_url = (known after apply) + svn_url = (known after apply) + topics = (known after apply) + visibility = "public" + vulnerability_alerts = (known after apply) + web_commit_signoff_required = (known after apply) + security_and_analysis (known after apply) } # github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key must be replaced -/+ resource "github_repository_deploy_key" "cmu-eats_dining_api_mirror_deploy_key" { ~ etag = "W/\"bf394a0fcc9e1231b4ee544598e6680359d71d177764d9f0c8cd22f5c7ec3ef4\"" -> (known after apply) ~ id = "dining-api:161772815" -> (known after apply) ~ key = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIIKyjdH3wH8J3aSgeZMLB9/uUGzyYltQsk5Cl1Q8MXqF" # forces replacement -> (known after apply) # forces replacement # (3 unchanged attributes hidden) } # github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key must be replaced -/+ resource "github_repository_deploy_key" "cmu-eats_eats_mirror_deploy_key" { ~ etag = "W/\"5cf445e3b52f1b2e59fe1cb32c5c2e0859e2598566a7aa75c3ebdbf4b01f6164\"" -> (known after apply) ~ id = "eats:161775995" -> (known after apply) ~ key = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFsm79Zlyi0JKH4y/BdgCCoLyrC4/18yF3owfsqOt53/" # forces replacement -> (known after apply) # forces replacement # (3 unchanged attributes hidden) } # github_repository_deploy_key.cmu-study_study_mirror_deploy_key will be created + resource "github_repository_deploy_key" "cmu-study_study_mirror_deploy_key" { + etag = (known after apply) + id = (known after apply) + key = (known after apply) + read_only = false + repository = "study" + title = "Forgejo mirror" } # github_repository_ruleset.cmu-eats_dining_api_ruleset will be updated in-place ~ resource "github_repository_ruleset" "cmu-eats_dining_api_ruleset" { id = "21879626" name = "Default" # (6 unchanged attributes hidden) ~ bypass_actors { ~ actor_id = 161772815 -> (known after apply) # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place ~ resource "github_repository_ruleset" "cmu-eats_eats_ruleset" { id = "21330218" name = "Default" # (6 unchanged attributes hidden) ~ bypass_actors { ~ actor_id = 161775995 -> (known after apply) # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.cmu-study_study_ruleset will be created + resource "github_repository_ruleset" "cmu-study_study_ruleset" { + enforcement = "active" + etag = (known after apply) + id = (known after apply) + name = "Default" + node_id = (known after apply) + repository = "study" + ruleset_id = (known after apply) + target = "branch" + bypass_actors { + actor_id = (known after apply) + actor_type = "DeployKey" + bypass_mode = "always" } + conditions { + ref_name { + exclude = [] + include = [ + "~ALL", ] } } + rules { + creation = true + deletion = true + non_fast_forward = true + update = true + update_allows_fetch_and_merge = false + pull_request { + allowed_merge_methods = (known after apply) + dismiss_stale_reviews_on_push = false + require_code_owner_review = false + require_last_push_approval = false + required_approving_review_count = 1 + required_review_thread_resolution = false } } } # github_team.cmu-study will be created + resource "github_team" "cmu-study" { + create_default_maintainer = false + etag = (known after apply) + id = (known after apply) + members_count = (known after apply) + name = "CMU Study" + node_id = (known after apply) + notification_setting = "notifications_enabled" + parent_team_read_id = (known after apply) + parent_team_read_slug = (known after apply) + privacy = "closed" + slug = (known after apply) } # github_team_membership.cmu-study_annadavi will be created + resource "github_team_membership" "cmu-study_annadavi" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = (known after apply) + username = "annadavid1" } # restapi_object.cmu-eats_dining_api_push_mirror will be created + resource "restapi_object" "cmu-eats_dining_api_push_mirror" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = jsonencode( { + interval = "8h0m0s" + remote_address = "git@github.com:ScottyLabs/dining-api.git" + sync_on_commit = true + use_ssh = true } ) + id = (known after apply) + id_attribute = "remote_name" + path = "/api/v1/repos/ScottyLabs/dining-api/push_mirrors" } # restapi_object.cmu-eats_eats_kennel_webhook will be created + resource "restapi_object" "cmu-eats_eats_kennel_webhook" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = (sensitive value) + id = (known after apply) + id_attribute = "id" + path = "/api/v1/repos/ScottyLabs/eats/hooks" + update_method = "PATCH" } # restapi_object.cmu-eats_eats_push_mirror will be created + resource "restapi_object" "cmu-eats_eats_push_mirror" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = jsonencode( { + interval = "8h0m0s" + remote_address = "git@github.com:ScottyLabs/eats.git" + sync_on_commit = true + use_ssh = true } ) + id = (known after apply) + id_attribute = "remote_name" + path = "/api/v1/repos/ScottyLabs/eats/push_mirrors" } # restapi_object.cmu-study_study_kennel_webhook will be created + resource "restapi_object" "cmu-study_study_kennel_webhook" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = (sensitive value) + id = (known after apply) + id_attribute = "id" + path = "/api/v1/repos/ScottyLabs/study/hooks" + update_method = "PATCH" } # restapi_object.cmu-study_study_push_mirror will be created + resource "restapi_object" "cmu-study_study_push_mirror" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = jsonencode( { + interval = "8h0m0s" + remote_address = "git@github.com:ScottyLabs/study.git" + sync_on_commit = true + use_ssh = true } ) + id = (known after apply) + id_attribute = "remote_name" + path = "/api/v1/repos/ScottyLabs/study/push_mirrors" } Plan: 20 to add, 4 to change, 2 to destroy. ╷ │ Error: Unable to read team repository │ │ with forgejo_team_repository.cmu-eats_dining_api, │ on forgejo_team_repos.tf.json line 102, in resource.forgejo_team_repository.cmu-eats_dining_api: │ 102: }, │ │ Repository 'ScottyLabs/dining-api' not found in team with ID 28 ╵ ╷ │ Error: Unable to read team repository │ │ with forgejo_team_repository.cmu-eats_eats, │ on forgejo_team_repos.tf.json line 111, in resource.forgejo_team_repository.cmu-eats_eats: │ 111: }, │ │ Repository 'ScottyLabs/eats' not found in team with ID 28 ╵ ``` </details> --- ### Plan Summary 10 projects, 8 with changes, 1 with no changes, 1 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis state -p git rm forgejo_team_repository.cmu-eats_dining_api forgejo_team_repository.cmu-eats_eats

atlantis state -p git rm forgejo_team_repository.cmu-eats_dining_api forgejo_team_repository.cmu-eats_eats

Ran State rm for project: git dir: tofu/git workspace: default

Removed forgejo_team_repository.cmu-eats_dining_api
Removed forgejo_team_repository.cmu-eats_eats
Successfully removed 2 resource instance(s).

🚮 A plan file was discarded. Re-plan would be required before applying.

  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    
Ran State `rm` for project: `git` dir: `tofu/git` workspace: `default` ```diff Removed forgejo_team_repository.cmu-eats_dining_api Removed forgejo_team_repository.cmu-eats_eats Successfully removed 2 resource instance(s). ``` :put_litter_in_its_place: A plan file was discarded. Re-plan would be required before applying. * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ```
Owner

atlantis plan -p git

atlantis plan -p git

Ran Plan for dir: tofu/git workspace: default

Plan Error

Show Output
running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu plan -input=false -refresh -out /var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git/git-default.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git': exit status 1
    Finished `release` profile [optimized] target(s) in 0.05s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
random_password.kennel_webhook_secret: Refreshing state... [id=none]
data.external.identity_gostmeaper: Reading...
data.external.identity_rhsen: Reading...
data.external.identity_jw8: Reading...
data.external.identity_qianxuege: Reading...
data.external.identity_liamwalk: Reading...
data.external.identity_acchs: Reading...
data.external.identity_kechezon: Reading...
data.external.identity_liamwalk: Read complete after 0s [id=-]
data.external.identity_jessica_lr_song: Reading...
data.external.identity_rhsen: Read complete after 0s [id=-]
data.external.identity_gostmeaper: Read complete after 0s [id=-]
data.external.identity_annadavi: Reading...
data.external.identity_zoexue: Reading...
data.external.identity_kechezon: Read complete after 0s [id=-]
data.external.identity_jiyas: Reading...
data.external.identity_qianxuege: Read complete after 0s [id=-]
data.external.identity_kothurvidur: Reading...
data.external.identity_jw8: Read complete after 0s [id=-]
data.external.identity_thesuperrl: Reading...
data.external.identity_acchs: Read complete after 0s [id=-]
data.external.identity_nikhilr: Reading...
data.external.identity_anish: Reading...
data.external.identity_amzoeee: Reading...
data.external.identity_stringatlas: Reading...
data.external.identity_jessica_lr_song: Read complete after 1s [id=-]
data.external.identity_krishsax: Reading...
data.external.identity_zoexue: Read complete after 1s [id=-]
data.external.identity_kothurvidur: Read complete after 1s [id=-]
data.external.identity_etashjha: Reading...
data.external.identity_arom1a: Reading...
data.external.identity_jiyas: Read complete after 1s [id=-]
data.external.identity_kluoxx: Reading...
data.external.identity_annadavi: Read complete after 1s [id=-]
data.external.identity_siddharthr: Reading...
data.external.identity_thesuperrl: Read complete after 1s [id=-]
data.external.identity_kritdass: Reading...
data.external.identity_nikhilr: Read complete after 1s [id=-]
data.external.identity_tentype: Reading...
data.external.identity_anish: Read complete after 1s [id=-]
data.external.identity_brightz: Reading...
data.external.identity_amzoeee: Read complete after 1s [id=-]
data.external.identity_rzhu3: Reading...
data.external.identity_stringatlas: Read complete after 1s [id=-]
data.external.identity_gabrielx: Reading...
data.external.identity_krishsax: Read complete after 0s [id=-]
data.external.identity_dkou: Reading...
data.external.identity_kluoxx: Read complete after 0s [id=-]
data.external.identity_jalenluorion: Reading...
data.external.identity_arom1a: Read complete after 0s [id=-]
data.external.identity_xboxbedrock: Reading...
data.external.identity_etashjha: Read complete after 0s [id=-]
data.external.identity_manrajs: Reading...
data.external.identity_siddharthr: Read complete after 0s [id=-]
data.external.identity_maybe_yiyi: Reading...
data.external.identity_kritdass: Read complete after 0s [id=-]
data.external.identity_jc_965: Reading...
data.external.identity_tentype: Read complete after 0s [id=-]
data.external.identity_dcrodrig: Reading...
data.external.identity_brightz: Read complete after 0s [id=-]
data.external.identity_emmawong: Reading...
data.external.identity_rzhu3: Read complete after 0s [id=-]
data.external.identity_riztofu: Reading...
data.external.identity_gabrielx: Read complete after 0s [id=-]
github_repository.quest_quest: Refreshing state... [id=quest]
data.external.identity_xboxbedrock: Read complete after 0s [id=-]
data.external.identity_dkou: Read complete after 0s [id=-]
github_team.cmu-cal: Refreshing state... [id=15669112]
github_repository.uia_scottylabs_org: Refreshing state... [id=scottylabs-org]
data.external.identity_jalenluorion: Read complete after 0s [id=-]
github_repository.cmu-cal_cal: Refreshing state... [id=cal]
data.external.identity_manrajs: Read complete after 0s [id=-]
data.external.identity_maybe_yiyi: Read complete after 0s [id=-]
github_team.occ: Refreshing state... [id=19240310]
github_team.terrier: Refreshing state... [id=12028577]
data.external.identity_jc_965: Read complete after 0s [id=-]
github_repository.slai_mcp_server: Refreshing state... [id=mcp-server]
data.external.identity_dcrodrig: Read complete after 0s [id=-]
github_team.cmu-eats: Refreshing state... [id=16807600]
data.external.identity_emmawong: Read complete after 0s [id=-]
github_repository.tartan-vote_tartan_vote: Refreshing state... [id=tartan-vote]
data.external.identity_riztofu: Read complete after 0s [id=-]
github_team.cmu-housing: Refreshing state... [id=17937919]
github_repository.cmu-housing_housing: Refreshing state... [id=housing]
github_team.lost-and-found: Refreshing state... [id=4088608]
github_repository.cmu-eats_dining_api: Refreshing state... [id=dining-api]
github_team.uia: Refreshing state... [id=17529763]
github_repository.cmu-courses_courses: Refreshing state... [id=courses]
github_repository.cbp_groupme_mirror: Refreshing state... [id=groupme-mirror]
github_repository.cmu-courses_internet_archive: Refreshing state... [id=internet-archive]
github_repository.slai_cmugpt_surface: Refreshing state... [id=cmugpt-surface]
github_team.slai: Refreshing state... [id=17571328]
github_repository.uia_components: Refreshing state... [id=components]
github_team.cmu-insta: Refreshing state... [id=19240311]
github_repository.devops_kennel: Refreshing state... [id=kennel]
github_team.quest: Refreshing state... [id=17330197]
github_repository.devops_documentation: Refreshing state... [id=documentation]
github_repository.terrier_terrier: Refreshing state... [id=terrier]
github_repository.slai_cmugpt_agent: Refreshing state... [id=cmugpt-agent]
github_repository.devops_governance: Refreshing state... [id=governance]
github_team.cbp: Refreshing state... [id=17170099]
github_repository.cbp_dalmatian: Refreshing state... [id=dalmatian]
github_repository.slai_cmugpt_sms_surface: Refreshing state... [id=cmugpt-sms-surface]
github_repository.cmu-eats_eats: Refreshing state... [id=eats]
github_team.cmu-courses: Refreshing state... [id=3643907]
github_repository.devops_infrastructure: Refreshing state... [id=infrastructure]
github_repository.devops_link_shortener: Refreshing state... [id=link-shortener]
github_repository.devops_collie: Refreshing state... [id=collie]
github_repository.cbp_bus_sign: Refreshing state... [id=bus-sign]
github_repository.lost-and-found_lost_and_found_v2: Refreshing state... [id=lost-and-found-v2]
github_repository.cmu-insta_insta: Refreshing state... [id=insta]
github_repository.cbp_discord_verify: Refreshing state... [id=discord-verify]
github_team.devops: Refreshing state... [id=14706218]
github_team.tartan-vote: Refreshing state... [id=17701833]
github_repository.occ_occ: Refreshing state... [id=occ]
github_team.cmu-maps: Refreshing state... [id=10872360]
github_repository.cmu-maps_maps: Refreshing state... [id=maps]
forgejo_repository.cbp_bus_sign: Refreshing state... [name=bus-sign]
forgejo_repository.slai_cmugpt_surface: Refreshing state... [name=cmugpt-surface]
forgejo_repository.cbp_discord_verify: Refreshing state... [name=discord-verify]
forgejo_repository.devops_collie: Refreshing state... [name=collie]
forgejo_repository.uia_scottylabs_org: Refreshing state... [name=scottylabs-org]
forgejo_repository.cmu-housing_housing: Refreshing state... [name=housing]
forgejo_repository.cmu-courses_courses: Refreshing state... [name=courses]
forgejo_repository.cbp_groupme_mirror: Refreshing state... [name=groupme-mirror]
forgejo_repository.devops_kennel: Refreshing state... [name=kennel]
forgejo_repository.slai_cmugpt_sms_surface: Refreshing state... [name=cmugpt-sms-surface]
forgejo_repository.cmu-cal_cal: Refreshing state... [name=cal]
forgejo_repository.uia_components: Refreshing state... [name=components]
forgejo_repository.cbp_dalmatian: Refreshing state... [name=dalmatian]
forgejo_repository.lost-and-found_lost_and_found_v2: Refreshing state... [name=lost-and-found-v2]
forgejo_repository.cmu-maps_maps: Refreshing state... [name=maps]
forgejo_repository.cmu-insta_insta: Refreshing state... [name=insta]
forgejo_repository.tartan-vote_tartan_vote: Refreshing state... [name=tartan-vote]
forgejo_repository.devops_governance: Refreshing state... [name=governance]
forgejo_repository.cmu-eats_eats: Refreshing state... [name=eats]
forgejo_repository.cmu-eats_dining_api: Refreshing state... [name=dining-api]
forgejo_repository.occ_occ: Refreshing state... [name=occ]
forgejo_repository.devops_documentation: Refreshing state... [name=documentation]
forgejo_repository.terrier_terrier: Refreshing state... [name=terrier]
forgejo_repository.quest_quest: Refreshing state... [name=quest]
forgejo_repository.devops_link_shortener: Refreshing state... [name=link-shortener]
forgejo_repository.slai_mcp_server: Refreshing state... [name=mcp-server]
forgejo_repository.slai_cmugpt_agent: Refreshing state... [name=cmugpt-agent]
data.forgejo_organization.this: Reading...
forgejo_repository.devops_infrastructure: Refreshing state... [name=infrastructure]
data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs]
forgejo_repository.cmu-courses_internet_archive: Refreshing state... [name=internet-archive]
github_team_membership.cmu-cal_jalenluorion: Refreshing state... [id=15669112:jalenluorion]
github_team_membership.cmu-cal_anish: Refreshing state... [id=15669112:ap-1]
github_team_membership.occ_xboxbedrock: Refreshing state... [id=19240310:xboxbedrock]
github_team_membership.occ_dcrodrig: Refreshing state... [id=19240310:newhacker1746]
github_team_membership.occ_acchs: Refreshing state... [id=19240310:acchs-at-cmu]
github_team_membership.occ_anish: Refreshing state... [id=19240310:ap-1]
github_team_membership.terrier_etashjha: Refreshing state... [id=12028577:etashj]
github_team_membership.terrier_kothurvidur: Refreshing state... [id=12028577:kothurvidur]
github_team_membership.terrier_dkou: Refreshing state... [id=12028577:sharkree]
github_team_membership.terrier_amzoeee: Refreshing state... [id=12028577:amzoeee]
github_team_membership.terrier_anish: Refreshing state... [id=12028577:ap-1]
github_team_membership.terrier_kritdass: Refreshing state... [id=12028577:kritdass]
github_team_membership.cmu-eats_arom1a: Refreshing state... [id=16807600:arom1a]
github_team_membership.cmu-housing_nikhilr: Refreshing state... [id=17937919:ecstaticpilot]
github_team_membership.cmu-housing_jessica_lr_song: Refreshing state... [id=17937919:jessica-lr-song]
github_team_membership.cmu-housing_manrajs: Refreshing state... [id=17937919:manraj10]
github_team_membership.cmu-housing_stringatlas: Refreshing state... [id=17937919:stringatlas]
github_team_membership.cmu-housing_xboxbedrock: Refreshing state... [id=17937919:xboxbedrock]
github_team_membership.cmu-housing_gostmeaper: Refreshing state... [id=17937919:gostmeaper]
github_team_membership.cmu-housing_amzoeee: Refreshing state... [id=17937919:amzoeee]
github_team_membership.cmu-housing_emmawong: Refreshing state... [id=17937919:emw868]
github_team_membership.cmu-housing_rhsen: Refreshing state... [id=17937919:rohansen08]
github_team_membership.cmu-housing_zoexue: Refreshing state... [id=17937919:o-zox-o]
github_team_membership.cmu-housing_kluoxx: Refreshing state... [id=17937919:luokaden888]
github_team_membership.cmu-housing_anish: Refreshing state... [id=17937919:ap-1]
github_team_membership.cmu-housing_tentype: Refreshing state... [id=17937919:tentype]
github_actions_repository_permissions.quest_quest_actions: Refreshing state... [id=quest]
github_actions_repository_permissions.uia_scottylabs_org_actions: Refreshing state... [id=scottylabs-org]
github_actions_repository_permissions.cmu-cal_cal_actions: Refreshing state... [id=cal]
github_team_membership.lost-and-found_anish: Refreshing state... [id=4088608:ap-1]
github_actions_repository_permissions.slai_mcp_server_actions: Refreshing state... [id=mcp-server]
github_team_membership.uia_anish: Refreshing state... [id=17529763:ap-1]
github_actions_repository_permissions.tartan-vote_tartan_vote_actions: Refreshing state... [id=tartan-vote]
github_actions_repository_permissions.cmu-courses_courses_actions: Refreshing state... [id=courses]
github_actions_repository_permissions.cbp_groupme_mirror_actions: Refreshing state... [id=groupme-mirror]
github_actions_repository_permissions.cmu-housing_housing_actions: Refreshing state... [id=housing]
github_actions_repository_permissions.cmu-courses_internet_archive_actions: Refreshing state... [id=internet-archive]
github_actions_repository_permissions.slai_cmugpt_surface_actions: Refreshing state... [id=cmugpt-surface]
github_team_membership.slai_krishsax: Refreshing state... [id=17571328:krishsaxena]
github_team_membership.slai_xboxbedrock: Refreshing state... [id=17571328:xboxbedrock]
github_team_membership.slai_qianxuege: Refreshing state... [id=17571328:qianxuege]
github_team_membership.slai_rzhu3: Refreshing state... [id=17571328:richard-31415]
github_team_membership.slai_jc_965: Refreshing state... [id=17571328:jc-965]
github_team_membership.slai_anish: Refreshing state... [id=17571328:ap-1]
github_team_membership.slai_jiyas: Refreshing state... [id=17571328:jiya2301]
github_actions_repository_permissions.cmu-eats_dining_api_actions: Refreshing state... [id=dining-api]
github_actions_repository_permissions.uia_components_actions: Refreshing state... [id=components]
github_team_membership.cmu-insta_etashjha: Refreshing state... [id=19240311:etashj]
github_team_membership.quest_kechezon: Refreshing state... [id=17330197:kechezon]
github_team_membership.quest_brightz: Refreshing state... [id=17330197:bright-zheng]
github_team_membership.quest_xboxbedrock: Refreshing state... [id=17330197:xboxbedrock]
github_team_membership.quest_jw8: Refreshing state... [id=17330197:jeffreyyijiwang]
github_team_membership.quest_anish: Refreshing state... [id=17330197:ap-1]
github_actions_repository_permissions.slai_cmugpt_agent_actions: Refreshing state... [id=cmugpt-agent]
github_team_membership.cbp_maybe_yiyi: Refreshing state... [id=17170099:maybe-yiyi]
github_team_membership.cbp_liamwalk: Refreshing state... [id=17170099:pandaaured]
github_team_membership.cbp_nikhilr: Refreshing state... [id=17170099:ecstaticpilot]
github_team_membership.cbp_xboxbedrock: Refreshing state... [id=17170099:xboxbedrock]
github_team_membership.cbp_anish: Refreshing state... [id=17170099:ap-1]
github_team_membership.cbp_tentype: Refreshing state... [id=17170099:tentype]
github_actions_repository_permissions.slai_cmugpt_sms_surface_actions: Refreshing state... [id=cmugpt-sms-surface]
github_actions_repository_permissions.devops_kennel_actions: Refreshing state... [id=kennel]
github_team_membership.cmu-courses_etashjha: Refreshing state... [id=3643907:etashj]
github_team_membership.cmu-courses_dkou: Refreshing state... [id=3643907:sharkree]
github_team_membership.cmu-courses_anish: Refreshing state... [id=3643907:ap-1]
github_team_membership.cmu-courses_gabrielx: Refreshing state... [id=3643907:gabriel-xu]
github_actions_repository_permissions.devops_documentation_actions: Refreshing state... [id=documentation]
github_actions_repository_permissions.terrier_terrier_actions: Refreshing state... [id=terrier]
github_actions_repository_permissions.devops_governance_actions: Refreshing state... [id=governance]
github_actions_repository_permissions.cbp_dalmatian_actions: Refreshing state... [id=dalmatian]
github_actions_repository_permissions.devops_collie_actions: Refreshing state... [id=collie]
github_actions_repository_permissions.cmu-eats_eats_actions: Refreshing state... [id=eats]
github_actions_repository_permissions.lost-and-found_lost_and_found_v2_actions: Refreshing state... [id=lost-and-found-v2]
github_actions_repository_permissions.cbp_discord_verify_actions: Refreshing state... [id=discord-verify]
restapi_object.cbp_bus_sign_kennel_webhook: Refreshing state... [id=21]
restapi_object.cbp_bus_sign_push_mirror: Refreshing state... [id=remote_mirror_MkKxQ2vA9pc]
restapi_object.slai_cmugpt_surface_push_mirror: Refreshing state... [id=remote_mirror_cPTQN6S5SpB]
restapi_object.slai_cmugpt_surface_kennel_webhook: Refreshing state... [id=23]
restapi_object.cbp_discord_verify_kennel_webhook: Refreshing state... [id=18]
restapi_object.cbp_discord_verify_push_mirror: Refreshing state... [id=remote_mirror_3MV7Up6cY1X]
restapi_object.devops_collie_push_mirror: Refreshing state... [id=remote_mirror_ljvVejNFXaS]
restapi_object.devops_collie_kennel_webhook: Refreshing state... [id=17]
restapi_object.uia_scottylabs_org_kennel_webhook: Refreshing state... [id=29]
restapi_object.uia_scottylabs_org_push_mirror: Refreshing state... [id=remote_mirror_h6yN6QdS1lD]
github_actions_repository_permissions.devops_infrastructure_actions: Refreshing state... [id=infrastructure]
restapi_object.cmu-housing_housing_kennel_webhook: Refreshing state... [id=16]
restapi_object.cmu-housing_housing_docs_webhook: Refreshing state... [id=15]
restapi_object.cmu-housing_housing_push_mirror: Refreshing state... [id=remote_mirror_uoV2HG4aiOT]
restapi_object.cmu-courses_courses_push_mirror: Refreshing state... [id=remote_mirror_WMR9yQwKFDF]
restapi_object.cmu-courses_courses_kennel_webhook: Refreshing state... [id=11]
restapi_object.cbp_groupme_mirror_push_mirror: Refreshing state... [id=remote_mirror_U61HJHB9VGE]
restapi_object.devops_kennel_kennel_webhook: Refreshing state... [id=14]
restapi_object.devops_kennel_push_mirror: Refreshing state... [id=remote_mirror_cT9psJkhTQ4]
restapi_object.slai_cmugpt_sms_surface_kennel_webhook: Refreshing state... [id=19]
restapi_object.slai_cmugpt_sms_surface_push_mirror: Refreshing state... [id=remote_mirror__fkBkASN2jA]
restapi_object.cmu-cal_cal_kennel_webhook: Refreshing state... [id=20]
restapi_object.cmu-cal_cal_push_mirror: Refreshing state... [id=remote_mirror_CO243uIONoq]
restapi_object.uia_components_push_mirror: Refreshing state... [id=remote_mirror_Kwwa8TDnTgc]
restapi_object.uia_components_kennel_webhook: Refreshing state... [id=27]
restapi_object.cbp_dalmatian_kennel_webhook: Refreshing state... [id=6]
restapi_object.cbp_dalmatian_docs_webhook: Refreshing state... [id=7]
restapi_object.cbp_dalmatian_push_mirror: Refreshing state... [id=remote_mirror_qzlCpLK153F]
restapi_object.lost-and-found_lost_and_found_v2_push_mirror: Refreshing state... [id=remote_mirror_n_UZBFU-CY4]
restapi_object.cmu-maps_maps_push_mirror: Refreshing state... [id=remote_mirror_e4_FUnvoslJ]
restapi_object.cmu-maps_maps_kennel_webhook: Refreshing state... [id=9]
github_team_membership.devops_anish: Refreshing state... [id=14706218:ap-1]
github_team_membership.devops_thesuperrl: Refreshing state... [id=14706218:thesuperrl]
github_team_membership.devops_xboxbedrock: Refreshing state... [id=14706218:xboxbedrock]
restapi_object.cmu-insta_insta_push_mirror: Refreshing state... [id=remote_mirror_891nu6XQ1us]
restapi_object.tartan-vote_tartan_vote_kennel_webhook: Refreshing state... [id=24]
restapi_object.tartan-vote_tartan_vote_push_mirror: Refreshing state... [id=remote_mirror_EUFq4sFSnH5]
restapi_object.tartan-vote_tartan_vote_docs_webhook: Refreshing state... [id=25]
restapi_object.devops_governance_push_mirror: Refreshing state... [id=remote_mirror_lt9JzkX1jMq]
restapi_object.devops_governance_kennel_webhook: Refreshing state... [id=22]
restapi_object.cmu-eats_eats_push_mirror: Refreshing state... [id=remote_mirror_sjYwMA6NPgh]
restapi_object.cmu-eats_eats_kennel_webhook: Refreshing state... [id=31]
restapi_object.cmu-eats_dining_api_push_mirror: Refreshing state... [id=remote_mirror_a4MsQ5rO1_Z]
restapi_object.occ_occ_push_mirror: Refreshing state... [id=remote_mirror_moBLq1YG1qz]
restapi_object.devops_documentation_push_mirror: Refreshing state... [id=remote_mirror_izmsqaNB6Wx]
restapi_object.terrier_terrier_push_mirror: Refreshing state... [id=remote_mirror_YopDgZksZv5]
restapi_object.terrier_terrier_kennel_webhook: Refreshing state... [id=12]
github_actions_repository_permissions.devops_link_shortener_actions: Refreshing state... [id=link-shortener]
restapi_object.quest_quest_push_mirror: Refreshing state... [id=remote_mirror_U9d9BPAon29]
restapi_object.quest_quest_kennel_webhook: Refreshing state... [id=10]
restapi_object.devops_link_shortener_push_mirror: Refreshing state... [id=remote_mirror_cp8daMpt82v]
restapi_object.devops_link_shortener_kennel_webhook: Refreshing state... [id=8]
restapi_object.slai_mcp_server_push_mirror: Refreshing state... [id=remote_mirror_Tqh2y0R_vDp]
restapi_object.slai_mcp_server_kennel_webhook: Refreshing state... [id=13]
forgejo_team.link-shortener_leads: Refreshing state... [name=link-shortener-leads]
forgejo_team.cmugpt-surface_leads: Refreshing state... [name=cmugpt-surface-leads]
forgejo_team.slai: Refreshing state... [name=slai]
forgejo_team.maps_leads: Refreshing state... [name=maps-leads]
forgejo_team.dalmatian_leads: Refreshing state... [name=dalmatian-leads]
forgejo_team.cmugpt-sms-surface_leads: Refreshing state... [name=cmugpt-sms-surface-leads]
forgejo_team.uia: Refreshing state... [name=uia]
forgejo_team.housing_leads: Refreshing state... [name=housing-leads]
forgejo_team.discord-verify_leads: Refreshing state... [name=discord-verify-leads]
forgejo_team.cmu-housing: Refreshing state... [name=cmu-housing]
forgejo_team.quest_leads: Refreshing state... [name=quest-leads]
forgejo_team.documentation_leads: Refreshing state... [name=documentation-leads]
forgejo_team.lost-and-found: Refreshing state... [name=lost-and-found]
forgejo_team.terrier_leads: Refreshing state... [name=terrier-leads]
forgejo_team.occ: Refreshing state... [name=occ]
forgejo_team.occ-main_leads: Refreshing state... [name=occ-main-leads]
forgejo_team.cbp: Refreshing state... [name=cbp]
forgejo_team.tartan-vote_leads: Refreshing state... [name=tartan-vote-leads]
forgejo_team.components_leads: Refreshing state... [name=components-leads]
forgejo_team.cmu-courses: Refreshing state... [name=cmu-courses]
forgejo_team.cmu-maps: Refreshing state... [name=cmu-maps]
forgejo_team.scottylabs-org_leads: Refreshing state... [name=scottylabs-org-leads]
forgejo_team.lost-and-found-v2_leads: Refreshing state... [name=lost-and-found-v2-leads]
forgejo_team.bus-sign_leads: Refreshing state... [name=bus-sign-leads]
forgejo_team.eats_leads: Refreshing state... [name=eats-leads]
forgejo_team.terrier: Refreshing state... [name=terrier]
forgejo_team.quest: Refreshing state... [name=quest]
forgejo_team.devops: Refreshing state... [name=devops]
forgejo_team.groupme-mirror_leads: Refreshing state... [name=groupme-mirror-leads]
forgejo_team.tartan-vote: Refreshing state... [name=tartan-vote]
forgejo_team.cmu-insta: Refreshing state... [name=cmu-insta]
forgejo_team.cmugpt-agent_leads: Refreshing state... [name=cmugpt-agent-leads]
forgejo_team.devops_leads: Refreshing state... [name=devops-leads]
forgejo_team.cmugpt-mcp-server_leads: Refreshing state... [name=cmugpt-mcp-server-leads]
forgejo_team.cmu-courses_leads: Refreshing state... [name=cmu-courses-leads]
forgejo_team.insta_leads: Refreshing state... [name=insta-leads]
forgejo_team.cmu-cal: Refreshing state... [name=cmu-cal]
forgejo_team.cal_leads: Refreshing state... [name=cal-leads]
forgejo_team.courses_leads: Refreshing state... [name=courses-leads]
forgejo_team.cmu-eats: Refreshing state... [name=cmu-eats]
restapi_object.slai_cmugpt_agent_push_mirror: Refreshing state... [id=remote_mirror_kNY6rVcxdyr]
restapi_object.slai_cmugpt_agent_kennel_webhook: Refreshing state... [id=28]
restapi_object.devops_infrastructure_push_mirror: Refreshing state... [id=remote_mirror_oYtFpKVzTaZ]
restapi_object.devops_infrastructure_kennel_webhook: Refreshing state... [id=26]
restapi_object.cmu-courses_internet_archive_push_mirror: Refreshing state... [id=remote_mirror_58X__Yfs0bR]
github_team_membership.tartan-vote_anish: Refreshing state... [id=17701833:ap-1]
github_team_membership.tartan-vote_maybe_yiyi: Refreshing state... [id=17701833:maybe-yiyi]
github_team_membership.tartan-vote_riztofu: Refreshing state... [id=17701833:riztofu]
github_team_membership.tartan-vote_tentype: Refreshing state... [id=17701833:tentype]
github_team_membership.tartan-vote_xboxbedrock: Refreshing state... [id=17701833:xboxbedrock]
github_team_membership.cmu-maps_siddharthr: Refreshing state... [id=10872360:overthrowing]
github_team_membership.cmu-maps_anish: Refreshing state... [id=10872360:ap-1]
github_actions_repository_permissions.cbp_bus_sign_actions: Refreshing state... [id=bus-sign]
github_actions_repository_permissions.cmu-insta_insta_actions: Refreshing state... [id=insta]
github_actions_repository_permissions.occ_occ_actions: Refreshing state... [id=occ]
github_actions_repository_permissions.cmu-maps_maps_actions: Refreshing state... [id=maps]
github_repository_deploy_key.cbp_bus_sign_mirror_deploy_key: Refreshing state... [id=bus-sign:159665342]
github_repository_deploy_key.slai_cmugpt_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-surface:159665348]
github_repository_deploy_key.cbp_discord_verify_mirror_deploy_key: Refreshing state... [id=discord-verify:159665340]
github_repository_deploy_key.devops_collie_mirror_deploy_key: Refreshing state... [id=collie:159665332]
github_repository_deploy_key.uia_scottylabs_org_mirror_deploy_key: Refreshing state... [id=scottylabs-org:159665361]
github_repository_deploy_key.cmu-housing_housing_mirror_deploy_key: Refreshing state... [id=housing:159665369]
github_repository_deploy_key.cmu-courses_courses_mirror_deploy_key: Refreshing state... [id=courses:159665327]
github_repository_deploy_key.cbp_groupme_mirror_mirror_deploy_key: Refreshing state... [id=groupme-mirror:159665366]
github_repository_deploy_key.devops_kennel_mirror_deploy_key: Refreshing state... [id=kennel:159665328]
github_repository_deploy_key.slai_cmugpt_sms_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-sms-surface:159665365]
github_repository_deploy_key.cmu-cal_cal_mirror_deploy_key: Refreshing state... [id=cal:159665344]
github_repository_deploy_key.uia_components_mirror_deploy_key: Refreshing state... [id=components:159665353]
github_repository_deploy_key.cbp_dalmatian_mirror_deploy_key: Refreshing state... [id=dalmatian:159665367]
github_repository_deploy_key.lost-and-found_lost_and_found_v2_mirror_deploy_key: Refreshing state... [id=lost-and-found-v2:159665337]
github_repository_deploy_key.cmu-maps_maps_mirror_deploy_key: Refreshing state... [id=maps:159665325]
github_repository_deploy_key.cmu-insta_insta_mirror_deploy_key: Refreshing state... [id=insta:161720245]
github_repository_deploy_key.tartan-vote_tartan_vote_mirror_deploy_key: Refreshing state... [id=tartan-vote:159665350]
github_repository_deploy_key.devops_governance_mirror_deploy_key: Refreshing state... [id=governance:159665346]
github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Refreshing state... [id=eats:161775995]
github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Refreshing state... [id=dining-api:161772815]
github_repository_deploy_key.occ_occ_mirror_deploy_key: Refreshing state... [id=occ:161720242]
github_repository_deploy_key.devops_documentation_mirror_deploy_key: Refreshing state... [id=documentation:159665362]
github_repository_deploy_key.terrier_terrier_mirror_deploy_key: Refreshing state... [id=terrier:159665339]
github_repository_deploy_key.quest_quest_mirror_deploy_key: Refreshing state... [id=quest:159665323]
github_repository_deploy_key.devops_link_shortener_mirror_deploy_key: Refreshing state... [id=link-shortener:159665345]
github_repository_deploy_key.slai_mcp_server_mirror_deploy_key: Refreshing state... [id=mcp-server:159665358]
forgejo_team_member.cmugpt-surface_leads_anish: Refreshing state...
forgejo_team_member.cmugpt-surface_leads_krishsax: Refreshing state...
forgejo_team_member.link-shortener_leads_anish: Refreshing state...
forgejo_team_member.link-shortener_leads_thesuperrl: Refreshing state...
forgejo_team_repository.slai_cmugpt_agent: Refreshing state...
forgejo_team_member.slai_jiyas: Refreshing state...
forgejo_team_repository.slai_cmugpt_surface: Refreshing state...
forgejo_team_member.slai_rzhu3: Refreshing state...
forgejo_team_repository.slai_cmugpt_sms_surface: Refreshing state...
forgejo_team_member.slai_qianxuege: Refreshing state...
forgejo_team_repository.slai_mcp_server: Refreshing state...
forgejo_team_member.slai_anish: Refreshing state...
forgejo_team_member.slai_jc_965: Refreshing state...
forgejo_team_member.slai_xboxbedrock: Refreshing state...
forgejo_team_member.slai_krishsax: Refreshing state...
forgejo_team_member.maps_leads_siddharthr: Refreshing state...
forgejo_team_member.maps_leads_anish: Refreshing state...
forgejo_team_member.dalmatian_leads_anish: Refreshing state...
forgejo_team_member.dalmatian_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.dalmatian_leads_tentype: Refreshing state...
forgejo_team_member.cmugpt-sms-surface_leads_krishsax: Refreshing state...
forgejo_team_member.cmugpt-sms-surface_leads_anish: Refreshing state...
forgejo_team_member.uia_anish: Refreshing state...
forgejo_team_repository.uia_scottylabs_org: Refreshing state...
forgejo_team_repository.uia_components: Refreshing state...
forgejo_team_member.housing_leads_anish: Refreshing state...
forgejo_team_member.housing_leads_gostmeaper: Refreshing state...
forgejo_team_member.housing_leads_nikhilr: Refreshing state...
forgejo_team_member.housing_leads_tentype: Refreshing state...
forgejo_team_member.housing_leads_emmawong: Refreshing state...
forgejo_team_member.discord-verify_leads_anish: Refreshing state...
forgejo_team_member.discord-verify_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.discord-verify_leads_tentype: Refreshing state...
forgejo_team_member.cmu-housing_stringatlas: Refreshing state...
forgejo_team_member.cmu-housing_zoexue: Refreshing state...
forgejo_team_member.cmu-housing_rhsen: Refreshing state...
forgejo_team_member.cmu-housing_gostmeaper: Refreshing state...
forgejo_team_member.cmu-housing_xboxbedrock: Refreshing state...
forgejo_team_repository.cmu-housing_housing: Refreshing state...
forgejo_team_member.cmu-housing_tentype: Refreshing state...
forgejo_team_member.cmu-housing_emmawong: Refreshing state...
forgejo_team_member.cmu-housing_kluoxx: Refreshing state...
forgejo_team_member.cmu-housing_amzoeee: Refreshing state...
forgejo_team_member.cmu-housing_nikhilr: Refreshing state...
forgejo_team_member.cmu-housing_anish: Refreshing state...
forgejo_team_member.cmu-housing_manrajs: Refreshing state...
forgejo_team_member.cmu-housing_jessica_lr_song: Refreshing state...
forgejo_team_member.quest_leads_anish: Refreshing state...
forgejo_team_member.quest_leads_xboxbedrock: Refreshing state...
forgejo_team_member.documentation_leads_thesuperrl: Refreshing state...
forgejo_team_member.documentation_leads_anish: Refreshing state...
forgejo_team_member.lost-and-found_anish: Refreshing state...
forgejo_team_repository.lost-and-found_lost_and_found_v2: Refreshing state...
forgejo_team_member.terrier_leads_anish: Refreshing state...
forgejo_team_member.occ_xboxbedrock: Refreshing state...
forgejo_team_member.terrier_leads_kritdass: Refreshing state...
forgejo_team_member.occ_acchs: Refreshing state...
forgejo_team_repository.occ_occ: Refreshing state...
forgejo_team_member.occ_dcrodrig: Refreshing state...
forgejo_team_member.occ_anish: Refreshing state...
forgejo_team_member.occ-main_leads_dcrodrig: Refreshing state...
forgejo_team_member.occ-main_leads_anish: Refreshing state...
forgejo_team_repository.cbp_bus_sign: Refreshing state...
forgejo_team_repository.cbp_discord_verify: Refreshing state...
forgejo_team_repository.cbp_dalmatian: Refreshing state...
forgejo_team_member.cbp_maybe_yiyi: Refreshing state...
forgejo_team_member.cbp_tentype: Refreshing state...
forgejo_team_member.cbp_anish: Refreshing state...
forgejo_team_member.cbp_liamwalk: Refreshing state...
forgejo_team_member.cbp_xboxbedrock: Refreshing state...
forgejo_team_repository.cbp_groupme_mirror: Refreshing state...
forgejo_team_member.cbp_nikhilr: Refreshing state...
forgejo_team_member.tartan-vote_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.tartan-vote_leads_anish: Refreshing state...
forgejo_team_member.components_leads_anish: Refreshing state...
forgejo_team_member.cmu-courses_etashjha: Refreshing state...
forgejo_team_member.cmu-courses_dkou: Refreshing state...
forgejo_team_repository.cmu-courses_courses: Refreshing state...
forgejo_team_member.cmu-courses_gabrielx: Refreshing state...
forgejo_team_repository.cmu-courses_internet_archive: Refreshing state...
forgejo_team_member.cmu-courses_anish: Refreshing state...
forgejo_team_repository.cmu-maps_maps: Refreshing state...
forgejo_team_member.cmu-maps_anish: Refreshing state...
forgejo_team_member.cmu-maps_siddharthr: Refreshing state...
forgejo_team_member.scottylabs-org_leads_anish: Refreshing state...
forgejo_team_member.lost-and-found-v2_leads_anish: Refreshing state...
forgejo_team_member.bus-sign_leads_anish: Refreshing state...
forgejo_team_member.bus-sign_leads_tentype: Refreshing state...
forgejo_team_member.bus-sign_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.terrier_kritdass: Refreshing state...
forgejo_team_member.terrier_dkou: Refreshing state...
forgejo_team_member.terrier_kothurvidur: Refreshing state...
forgejo_team_member.terrier_amzoeee: Refreshing state...
forgejo_team_member.terrier_etashjha: Refreshing state...
forgejo_team_member.terrier_anish: Refreshing state...
forgejo_team_repository.terrier_terrier: Refreshing state...
forgejo_team_member.eats_leads_arom1a: Refreshing state...
forgejo_team_member.quest_jw8: Refreshing state...
forgejo_team_member.quest_kechezon: Refreshing state...
forgejo_team_member.quest_brightz: Refreshing state...
forgejo_team_member.quest_anish: Refreshing state...
forgejo_team_member.quest_xboxbedrock: Refreshing state...
forgejo_team_repository.quest_quest: Refreshing state...
forgejo_team_repository.devops_kennel: Refreshing state...
forgejo_team_repository.devops_collie: Refreshing state...
forgejo_team_repository.devops_infrastructure: Refreshing state...
forgejo_team_member.devops_anish: Refreshing state...
forgejo_team_repository.devops_documentation: Refreshing state...
forgejo_team_member.devops_thesuperrl: Refreshing state...
forgejo_team_repository.devops_governance: Refreshing state...
forgejo_team_member.devops_xboxbedrock: Refreshing state...
forgejo_team_repository.devops_link_shortener: Refreshing state...
forgejo_team_member.groupme-mirror_leads_maybe_yiyi: Refreshing state...
forgejo_team_member.groupme-mirror_leads_tentype: Refreshing state...
forgejo_team_member.groupme-mirror_leads_anish: Refreshing state...
forgejo_team_member.tartan-vote_xboxbedrock: Refreshing state...
forgejo_team_repository.tartan-vote_tartan_vote: Refreshing state...
forgejo_team_member.tartan-vote_riztofu: Refreshing state...
forgejo_team_member.tartan-vote_anish: Refreshing state...
forgejo_team_member.tartan-vote_tentype: Refreshing state...
forgejo_team_member.tartan-vote_maybe_yiyi: Refreshing state...
forgejo_team_member.cmugpt-agent_leads_anish: Refreshing state...
forgejo_team_member.cmugpt-agent_leads_krishsax: Refreshing state...
forgejo_team_repository.cmu-insta_insta: Refreshing state...
forgejo_team_member.cmu-insta_etashjha: Refreshing state...
forgejo_team_member.devops_leads_thesuperrl: Refreshing state...
forgejo_team_member.devops_leads_anish: Refreshing state...
forgejo_team_member.cmugpt-mcp-server_leads_krishsax: Refreshing state...
forgejo_team_member.cmugpt-mcp-server_leads_anish: Refreshing state...
forgejo_team_member.cmu-courses_leads_anish: Refreshing state...
forgejo_team_member.cmu-courses_leads_etashjha: Refreshing state...
forgejo_team_member.insta_leads_etashjha: Refreshing state...
forgejo_team_member.cmu-cal_jalenluorion: Refreshing state...
forgejo_team_member.cmu-cal_anish: Refreshing state...
forgejo_team_member.cal_leads_jalenluorion: Refreshing state...
forgejo_team_repository.cmu-cal_cal: Refreshing state...
forgejo_team_member.cal_leads_anish: Refreshing state...
forgejo_team_member.courses_leads_anish: Refreshing state...
forgejo_team_member.cmu-eats_arom1a: Refreshing state...
forgejo_team_member.courses_leads_etashjha: Refreshing state...
github_repository_deploy_key.slai_cmugpt_agent_mirror_deploy_key: Refreshing state... [id=cmugpt-agent:159665360]
github_repository_deploy_key.devops_infrastructure_mirror_deploy_key: Refreshing state... [id=infrastructure:159665343]
github_repository_deploy_key.cmu-courses_internet_archive_mirror_deploy_key: Refreshing state... [id=internet-archive:159665356]
github_repository_ruleset.cbp_bus_sign_ruleset: Refreshing state... [id=15236550]
github_repository_ruleset.slai_cmugpt_surface_ruleset: Refreshing state... [id=16476662]
github_repository_ruleset.cbp_discord_verify_ruleset: Refreshing state... [id=11449154]
github_repository_ruleset.devops_collie_ruleset: Refreshing state... [id=18865204]
github_repository_ruleset.uia_scottylabs_org_ruleset: Refreshing state... [id=18112214]
github_repository_ruleset.cmu-housing_housing_ruleset: Refreshing state... [id=17421604]
github_repository_ruleset.cmu-courses_courses_ruleset: Refreshing state... [id=15855161]
github_repository_ruleset.cbp_groupme_mirror_ruleset: Refreshing state... [id=12503296]
github_repository_ruleset.devops_kennel_ruleset: Refreshing state... [id=13193453]
github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Refreshing state... [id=17653622]
github_repository_ruleset.cmu-cal_cal_ruleset: Refreshing state... [id=18919281]
github_repository_ruleset.uia_components_ruleset: Refreshing state... [id=16366085]
github_repository_ruleset.cbp_dalmatian_ruleset: Refreshing state... [id=17238915]
github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Refreshing state... [id=18920618]
github_repository_ruleset.cmu-maps_maps_ruleset: Refreshing state... [id=18864961]
github_repository_ruleset.cmu-insta_insta_ruleset: Refreshing state... [id=21841636]
github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Refreshing state... [id=16848966]
forgejo_team_repository.cmugpt-agent_leads_cmugpt_agent: Refreshing state...
forgejo_team_repository.cmugpt-surface_leads_cmugpt_surface: Refreshing state...
forgejo_team_repository.cmugpt-sms-surface_leads_cmugpt_sms_surface: Refreshing state...
forgejo_team_repository.cmugpt-mcp-server_leads_mcp_server: Refreshing state...
github_repository_ruleset.devops_governance_ruleset: Refreshing state... [id=14488792]
forgejo_team_repository.scottylabs-org_leads_scottylabs_org: Refreshing state...
forgejo_team_repository.components_leads_components: Refreshing state...
github_repository_ruleset.cmu-eats_eats_ruleset: Refreshing state... [id=21330218]
forgejo_team_repository.housing_leads_housing: Refreshing state...
forgejo_team_repository.lost-and-found-v2_leads_lost_and_found_v2: Refreshing state...
github_repository_ruleset.cmu-eats_dining_api_ruleset: Refreshing state... [id=21879626]
forgejo_team_repository.occ-main_leads_occ: Refreshing state...
forgejo_team_repository.bus-sign_leads_bus_sign: Refreshing state...
forgejo_team_repository.discord-verify_leads_discord_verify: Refreshing state...
forgejo_team_repository.dalmatian_leads_dalmatian: Refreshing state...
forgejo_team_repository.groupme-mirror_leads_groupme_mirror: Refreshing state...
github_repository_ruleset.occ_occ_ruleset: Refreshing state... [id=21841634]
forgejo_team_repository.courses_leads_courses: Refreshing state...
forgejo_team_repository.cmu-courses_leads_internet_archive: Refreshing state...
forgejo_team_repository.maps_leads_maps: Refreshing state...
forgejo_team_repository.terrier_leads_terrier: Refreshing state...
github_repository_ruleset.devops_documentation_ruleset: Refreshing state... [id=17079368]
forgejo_team_repository.quest_leads_quest: Refreshing state...
forgejo_team_repository.devops_leads_kennel: Refreshing state...
forgejo_team_repository.documentation_leads_documentation: Refreshing state...
forgejo_team_repository.devops_leads_collie: Refreshing state...
forgejo_team_repository.devops_leads_infrastructure: Refreshing state...
forgejo_team_repository.devops_leads_governance: Refreshing state...
forgejo_team_repository.link-shortener_leads_link_shortener: Refreshing state...
forgejo_team_repository.tartan-vote_leads_tartan_vote: Refreshing state...
forgejo_team_repository.insta_leads_insta: Refreshing state...
github_repository_ruleset.terrier_terrier_ruleset: Refreshing state... [id=12671003]
forgejo_team_repository.eats_leads_eats: Refreshing state...
forgejo_team_repository.eats_leads_dining_api: Refreshing state...
forgejo_team_repository.cal_leads_cal: Refreshing state...
github_repository_ruleset.quest_quest_ruleset: Refreshing state... [id=15750052]
github_repository_ruleset.devops_link_shortener_ruleset: Refreshing state... [id=17911647]
github_repository_ruleset.slai_mcp_server_ruleset: Refreshing state... [id=17082154]
github_repository_ruleset.slai_cmugpt_agent_ruleset: Refreshing state... [id=16476661]
github_repository_ruleset.devops_infrastructure_ruleset: Refreshing state... [id=11275118]
github_repository_ruleset.cmu-courses_internet_archive_ruleset: Refreshing state... [id=11988074]

Note: Objects have changed outside of OpenTofu

OpenTofu detected the following changes made outside of OpenTofu since the
last "tofu apply" which may have affected this plan:

  # restapi_object.cmu-eats_dining_api_push_mirror has been deleted
  - resource "restapi_object" "cmu-eats_dining_api_push_mirror" {
      - api_data        = {
          - "public_key"     = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIIKyjdH3wH8J3aSgeZMLB9/uUGzyYltQsk5Cl1Q8MXqF"
            # (9 unchanged elements hidden)
        } -> null
        id              = "remote_mirror_a4MsQ5rO1_Z"
        # (5 unchanged attributes hidden)
    }

  # restapi_object.cmu-eats_eats_push_mirror has been deleted
  - resource "restapi_object" "cmu-eats_eats_push_mirror" {
      - api_data        = {
          - "public_key"     = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFsm79Zlyi0JKH4y/BdgCCoLyrC4/18yF3owfsqOt53/"
            # (9 unchanged elements hidden)
        } -> null
        id              = "remote_mirror_sjYwMA6NPgh"
        # (5 unchanged attributes hidden)
    }


Unless you have made equivalent changes to your configuration, or ignored the
relevant attributes using ignore_changes, the following plan may include
actions to undo or respond to these changes.

─────────────────────────────────────────────────────────────────────────────

OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
  + create
  ~ update in-place (current -> planned)
-/+ destroy and then create replacement

OpenTofu planned the following actions, but then encountered a problem:

  # forgejo_repository.cmu-eats_dining_api will be updated in-place
  ~ resource "forgejo_repository" "cmu-eats_dining_api" {
      ~ clone_url                   = "https://git.cmu.dev/ScottyLabs/dining-api.git" -> (known after apply)
      ~ created_at                  = "2026-09-18T23:16:27Z" -> (known after apply)
      + description                 = "Dining API for CMU dining locations and availability"
      ~ empty                       = false -> (known after apply)
      + external_tracker            = (known after apply)
      + external_wiki               = (known after apply)
      ~ forks_count                 = 0 -> (known after apply)
      ~ full_name                   = "ScottyLabs/dining-api" -> (known after apply)
      ~ html_url                    = "https://git.cmu.dev/ScottyLabs/dining-api" -> (known after apply)
        id                          = 97
      ~ internal_tracker            = {
          ~ allow_only_contributors_to_track_time = true -> (known after apply)
          ~ enable_issue_dependencies             = true -> (known after apply)
          ~ enable_time_tracker                   = true -> (known after apply)
        } -> (known after apply)
      + mirror_interval             = (known after apply)
      ~ mirror_updated              = "0001-01-01T00:00:00Z" -> (known after apply)
        name                        = "dining-api"
      ~ open_issues_count           = 0 -> (known after apply)
      ~ open_pr_counter             = 5 -> (known after apply)
      ~ permissions                 = {
          ~ admin = true -> (known after apply)
          ~ pull  = true -> (known after apply)
          ~ push  = true -> (known after apply)
        } -> (known after apply)
      ~ release_counter             = 0 -> (known after apply)
      ~ size                        = 38816 -> (known after apply)
      ~ ssh_url                     = "ssh://forgejo@git.cmu.dev/ScottyLabs/dining-api.git" -> (known after apply)
      ~ stars_count                 = 0 -> (known after apply)
      ~ updated_at                  = "2026-09-18T23:22:22Z" -> (known after apply)
      ~ watchers_count              = 3 -> (known after apply)
      ~ website                     = "https://git.cmu.dev/ScottyLabs/dining-api" -> "https://api.cmueats.com"
        # (30 unchanged attributes hidden)
    }

  # forgejo_repository.cmu-eats_eats will be updated in-place
  ~ resource "forgejo_repository" "cmu-eats_eats" {
      ~ clone_url                   = "https://git.cmu.dev/ScottyLabs/eats.git" -> (known after apply)
      ~ created_at                  = "2026-09-18T23:15:39Z" -> (known after apply)
      ~ description                 = "Vite web app for CMU dining location info" -> "CMU dining location info"
      ~ empty                       = false -> (known after apply)
      + external_tracker            = (known after apply)
      + external_wiki               = (known after apply)
      ~ forks_count                 = 0 -> (known after apply)
      ~ full_name                   = "ScottyLabs/eats" -> (known after apply)
      ~ html_url                    = "https://git.cmu.dev/ScottyLabs/eats" -> (known after apply)
        id                          = 96
      ~ internal_tracker            = {
          ~ allow_only_contributors_to_track_time = true -> (known after apply)
          ~ enable_issue_dependencies             = true -> (known after apply)
          ~ enable_time_tracker                   = true -> (known after apply)
        } -> (known after apply)
      + mirror_interval             = (known after apply)
      ~ mirror_updated              = "0001-01-01T00:00:00Z" -> (known after apply)
        name                        = "eats"
      ~ open_issues_count           = 48 -> (known after apply)
      ~ open_pr_counter             = 4 -> (known after apply)
      ~ permissions                 = {
          ~ admin = true -> (known after apply)
          ~ pull  = true -> (known after apply)
          ~ push  = true -> (known after apply)
        } -> (known after apply)
      ~ release_counter             = 0 -> (known after apply)
      ~ size                        = 95496 -> (known after apply)
      ~ ssh_url                     = "ssh://forgejo@git.cmu.dev/ScottyLabs/eats.git" -> (known after apply)
      ~ stars_count                 = 0 -> (known after apply)
      ~ updated_at                  = "2026-09-18T23:35:12Z" -> (known after apply)
      ~ watchers_count              = 3 -> (known after apply)
        # (31 unchanged attributes hidden)
    }

  # forgejo_repository.cmu-study_study will be created
  + resource "forgejo_repository" "cmu-study_study" {
      + allow_manual_merge          = false
      + allow_merge_commits         = true
      + allow_rebase                = true
      + allow_rebase_explicit       = true
      + allow_squash_merge          = true
      + archive_on_destroy          = false
      + archived                    = false
      + auto_init                   = true
      + autodetect_manual_merge     = false
      + avatar_url                  = (known after apply)
      + clone_url                   = (known after apply)
      + created_at                  = (known after apply)
      + default_branch              = "main"
      + default_merge_style         = "merge"
      + description                 = "A platform for creating and joining study groups"
      + empty                       = (known after apply)
      + external_tracker            = (known after apply)
      + external_wiki               = (known after apply)
      + fork                        = (known after apply)
      + forks_count                 = (known after apply)
      + full_name                   = (known after apply)
      + has_actions                 = true
      + has_issues                  = true
      + has_packages                = true
      + has_projects                = true
      + has_pull_requests           = true
      + has_releases                = true
      + has_wiki                    = true
      + html_url                    = (known after apply)
      + id                          = (known after apply)
      + ignore_whitespace_conflicts = false
      + internal                    = (known after apply)
      + internal_tracker            = (known after apply)
      + labels                      = false
      + lfs                         = false
      + milestones                  = false
      + mirror                      = false
      + mirror_interval             = (known after apply)
      + mirror_updated              = (known after apply)
      + name                        = "study"
      + open_issues_count           = (known after apply)
      + open_pr_counter             = (known after apply)
      + owner                       = "ScottyLabs"
      + parent_id                   = (known after apply)
      + permissions                 = (known after apply)
      + private                     = false
      + release_counter             = (known after apply)
      + size                        = (known after apply)
      + ssh_url                     = (known after apply)
      + stars_count                 = (known after apply)
      + template                    = false
      + trust_model                 = "default"
      + updated_at                  = (known after apply)
      + watchers_count              = (known after apply)
      + website                     = "https://cmustudy.com"
    }

  # forgejo_team.cmu-study will be created
  + resource "forgejo_team" "cmu-study" {
      + can_create_org_repo       = (known after apply)
      + description               = "CMU Study"
      + id                        = (known after apply)
      + includes_all_repositories = (known after apply)
      + name                      = "cmu-study"
      + organization              = (known after apply)
      + organization_id           = 3
      + permission                = "write"
      + units_map                 = {
          + "repo.code"   = "write"
          + "repo.issues" = "write"
          + "repo.pulls"  = "write"
        }
    }

  # forgejo_team.study_leads will be created
  + resource "forgejo_team" "study_leads" {
      + can_create_org_repo       = false
      + description               = "CMU Study leads"
      + id                        = (known after apply)
      + includes_all_repositories = false
      + name                      = "study-leads"
      + organization              = (known after apply)
      + organization_id           = 3
      + permission                = "admin"
      + units_map                 = {
          + "repo.actions"    = "admin"
          + "repo.code"       = "admin"
          + "repo.ext_issues" = "admin"
          + "repo.ext_wiki"   = "admin"
          + "repo.issues"     = "admin"
          + "repo.packages"   = "admin"
          + "repo.projects"   = "admin"
          + "repo.pulls"      = "admin"
          + "repo.releases"   = "admin"
          + "repo.wiki"       = "admin"
        }
    }

  # forgejo_team_member.cmu-study_annadavi will be created
  + resource "forgejo_team_member" "cmu-study_annadavi" {
      + team_id = (known after apply)
      + user    = "annadavi"
    }

  # forgejo_team_member.study_leads_annadavi will be created
  + resource "forgejo_team_member" "study_leads_annadavi" {
      + team_id = (known after apply)
      + user    = "annadavi"
    }

  # forgejo_team_repository.cmu-eats_dining_api will be created
  + resource "forgejo_team_repository" "cmu-eats_dining_api" {
      + owner      = "ScottyLabs"
      + repository = "dining-api"
      + team_id    = 28
    }

  # forgejo_team_repository.cmu-eats_eats will be created
  + resource "forgejo_team_repository" "cmu-eats_eats" {
      + owner      = "ScottyLabs"
      + repository = "eats"
      + team_id    = 28
    }

  # forgejo_team_repository.cmu-study_study will be created
  + resource "forgejo_team_repository" "cmu-study_study" {
      + owner      = "ScottyLabs"
      + repository = "study"
      + team_id    = (known after apply)
    }

  # forgejo_team_repository.study_leads_study will be created
  + resource "forgejo_team_repository" "study_leads_study" {
      + owner      = "ScottyLabs"
      + repository = "study"
      + team_id    = (known after apply)
    }

  # github_actions_repository_permissions.cmu-study_study_actions will be created
  + resource "github_actions_repository_permissions" "cmu-study_study_actions" {
      + enabled              = false
      + id                   = (known after apply)
      + repository           = "study"
      + sha_pinning_required = (known after apply)
    }

  # github_repository.cmu-study_study will be created
  + resource "github_repository" "cmu-study_study" {
      + allow_auto_merge                        = false
      + allow_forking                           = (known after apply)
      + allow_merge_commit                      = true
      + allow_rebase_merge                      = true
      + allow_squash_merge                      = true
      + archived                                = false
      + default_branch                          = (known after apply)
      + delete_branch_on_merge                  = false
      + description                             = "READ-ONLY MIRROR"
      + etag                                    = (known after apply)
      + fork                                    = (known after apply)
      + full_name                               = (known after apply)
      + git_clone_url                           = (known after apply)
      + has_discussions                         = false
      + has_issues                              = false
      + has_projects                            = false
      + has_wiki                                = false
      + homepage_url                            = "https://git.cmu.dev/ScottyLabs/study"
      + html_url                                = (known after apply)
      + http_clone_url                          = (known after apply)
      + id                                      = (known after apply)
      + ignore_vulnerability_alerts_during_read = false
      + merge_commit_message                    = "PR_TITLE"
      + merge_commit_title                      = "MERGE_MESSAGE"
      + name                                    = "study"
      + node_id                                 = (known after apply)
      + primary_language                        = (known after apply)
      + private                                 = (known after apply)
      + repo_id                                 = (known after apply)
      + source_owner                            = (known after apply)
      + source_repo                             = (known after apply)
      + squash_merge_commit_message             = "COMMIT_MESSAGES"
      + squash_merge_commit_title               = "COMMIT_OR_PR_TITLE"
      + ssh_clone_url                           = (known after apply)
      + svn_url                                 = (known after apply)
      + topics                                  = (known after apply)
      + visibility                              = "public"
      + vulnerability_alerts                    = (known after apply)
      + web_commit_signoff_required             = (known after apply)

      + security_and_analysis (known after apply)
    }

  # github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key must be replaced
-/+ resource "github_repository_deploy_key" "cmu-eats_dining_api_mirror_deploy_key" {
      ~ etag       = "W/\"bf394a0fcc9e1231b4ee544598e6680359d71d177764d9f0c8cd22f5c7ec3ef4\"" -> (known after apply)
      ~ id         = "dining-api:161772815" -> (known after apply)
      ~ key        = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIIKyjdH3wH8J3aSgeZMLB9/uUGzyYltQsk5Cl1Q8MXqF" # forces replacement -> (known after apply) # forces replacement
        # (3 unchanged attributes hidden)
    }

  # github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key must be replaced
-/+ resource "github_repository_deploy_key" "cmu-eats_eats_mirror_deploy_key" {
      ~ etag       = "W/\"5cf445e3b52f1b2e59fe1cb32c5c2e0859e2598566a7aa75c3ebdbf4b01f6164\"" -> (known after apply)
      ~ id         = "eats:161775995" -> (known after apply)
      ~ key        = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFsm79Zlyi0JKH4y/BdgCCoLyrC4/18yF3owfsqOt53/" # forces replacement -> (known after apply) # forces replacement
        # (3 unchanged attributes hidden)
    }

  # github_repository_deploy_key.cmu-study_study_mirror_deploy_key will be created
  + resource "github_repository_deploy_key" "cmu-study_study_mirror_deploy_key" {
      + etag       = (known after apply)
      + id         = (known after apply)
      + key        = (known after apply)
      + read_only  = false
      + repository = "study"
      + title      = "Forgejo mirror"
    }

  # github_repository_ruleset.cmu-eats_dining_api_ruleset will be updated in-place
  ~ resource "github_repository_ruleset" "cmu-eats_dining_api_ruleset" {
        id          = "21879626"
        name        = "Default"
        # (6 unchanged attributes hidden)

      ~ bypass_actors {
          ~ actor_id    = 161772815 -> (known after apply)
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place
  ~ resource "github_repository_ruleset" "cmu-eats_eats_ruleset" {
        id          = "21330218"
        name        = "Default"
        # (6 unchanged attributes hidden)

      ~ bypass_actors {
          ~ actor_id    = 161775995 -> (known after apply)
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.cmu-study_study_ruleset will be created
  + resource "github_repository_ruleset" "cmu-study_study_ruleset" {
      + enforcement = "active"
      + etag        = (known after apply)
      + id          = (known after apply)
      + name        = "Default"
      + node_id     = (known after apply)
      + repository  = "study"
      + ruleset_id  = (known after apply)
      + target      = "branch"

      + bypass_actors {
          + actor_id    = (known after apply)
          + actor_type  = "DeployKey"
          + bypass_mode = "always"
        }

      + conditions {
          + ref_name {
              + exclude = []
              + include = [
                  + "~ALL",
                ]
            }
        }

      + rules {
          + creation                      = true
          + deletion                      = true
          + non_fast_forward              = true
          + update                        = true
          + update_allows_fetch_and_merge = false

          + pull_request {
              + allowed_merge_methods             = (known after apply)
              + dismiss_stale_reviews_on_push     = false
              + require_code_owner_review         = false
              + require_last_push_approval        = false
              + required_approving_review_count   = 1
              + required_review_thread_resolution = false
            }
        }
    }

  # github_team.cmu-study will be created
  + resource "github_team" "cmu-study" {
      + create_default_maintainer = false
      + etag                      = (known after apply)
      + id                        = (known after apply)
      + members_count             = (known after apply)
      + name                      = "CMU Study"
      + node_id                   = (known after apply)
      + notification_setting      = "notifications_enabled"
      + parent_team_read_id       = (known after apply)
      + parent_team_read_slug     = (known after apply)
      + privacy                   = "closed"
      + slug                      = (known after apply)
    }

  # github_team_membership.cmu-study_annadavi will be created
  + resource "github_team_membership" "cmu-study_annadavi" {
      + etag     = (known after apply)
      + id       = (known after apply)
      + role     = "maintainer"
      + team_id  = (known after apply)
      + username = "annadavid1"
    }

  # restapi_object.cmu-eats_dining_api_push_mirror will be created
  + resource "restapi_object" "cmu-eats_dining_api_push_mirror" {
      + api_data        = (known after apply)
      + api_response    = (known after apply)
      + create_response = (known after apply)
      + data            = jsonencode(
            {
              + interval       = "8h0m0s"
              + remote_address = "git@github.com:ScottyLabs/dining-api.git"
              + sync_on_commit = true
              + use_ssh        = true
            }
        )
      + id              = (known after apply)
      + id_attribute    = "remote_name"
      + path            = "/api/v1/repos/ScottyLabs/dining-api/push_mirrors"
    }

  # restapi_object.cmu-eats_eats_kennel_webhook will be created
  + resource "restapi_object" "cmu-eats_eats_kennel_webhook" {
      + api_data        = (known after apply)
      + api_response    = (known after apply)
      + create_response = (known after apply)
      + data            = (sensitive value)
      + id              = (known after apply)
      + id_attribute    = "id"
      + path            = "/api/v1/repos/ScottyLabs/eats/hooks"
      + update_method   = "PATCH"
    }

  # restapi_object.cmu-eats_eats_push_mirror will be created
  + resource "restapi_object" "cmu-eats_eats_push_mirror" {
      + api_data        = (known after apply)
      + api_response    = (known after apply)
      + create_response = (known after apply)
      + data            = jsonencode(
            {
              + interval       = "8h0m0s"
              + remote_address = "git@github.com:ScottyLabs/eats.git"
              + sync_on_commit = true
              + use_ssh        = true
            }
        )
      + id              = (known after apply)
      + id_attribute    = "remote_name"
      + path            = "/api/v1/repos/ScottyLabs/eats/push_mirrors"
    }

  # restapi_object.cmu-study_study_kennel_webhook will be created
  + resource "restapi_object" "cmu-study_study_kennel_webhook" {
      + api_data        = (known after apply)
      + api_response    = (known after apply)
      + create_response = (known after apply)
      + data            = (sensitive value)
      + id              = (known after apply)
      + id_attribute    = "id"
      + path            = "/api/v1/repos/ScottyLabs/study/hooks"
      + update_method   = "PATCH"
    }

  # restapi_object.cmu-study_study_push_mirror will be created
  + resource "restapi_object" "cmu-study_study_push_mirror" {
      + api_data        = (known after apply)
      + api_response    = (known after apply)
      + create_response = (known after apply)
      + data            = jsonencode(
            {
              + interval       = "8h0m0s"
              + remote_address = "git@github.com:ScottyLabs/study.git"
              + sync_on_commit = true
              + use_ssh        = true
            }
        )
      + id              = (known after apply)
      + id_attribute    = "remote_name"
      + path            = "/api/v1/repos/ScottyLabs/study/push_mirrors"
    }

Plan: 22 to add, 4 to change, 2 to destroy.
╷
│ Error: Unable to read team repository
│ 
│   with forgejo_team_repository.eats_leads_dining_api,
│   on forgejo_team_repos.tf.json line 327, in resource.forgejo_team_repository.eats_leads_dining_api:
│  327:       },
│ 
│ Repository 'ScottyLabs/dining-api' not found in team with ID 27
╵
╷
│ Error: Unable to read team repository
│ 
│   with forgejo_team_repository.eats_leads_eats,
│   on forgejo_team_repos.tf.json line 336, in resource.forgejo_team_repository.eats_leads_eats:
│  336:       },
│ 
│ Repository 'ScottyLabs/eats' not found in team with ID 27
╵


Ran Plan for dir: `tofu/git` workspace: `default` **Plan Error** <details><summary>Show Output</summary> ``` running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu plan -input=false -refresh -out /var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git/git-default.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git': exit status 1 Finished `release` profile [optimized] target(s) in 0.05s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ random_password.kennel_webhook_secret: Refreshing state... [id=none] data.external.identity_gostmeaper: Reading... data.external.identity_rhsen: Reading... data.external.identity_jw8: Reading... data.external.identity_qianxuege: Reading... data.external.identity_liamwalk: Reading... data.external.identity_acchs: Reading... data.external.identity_kechezon: Reading... data.external.identity_liamwalk: Read complete after 0s [id=-] data.external.identity_jessica_lr_song: Reading... data.external.identity_rhsen: Read complete after 0s [id=-] data.external.identity_gostmeaper: Read complete after 0s [id=-] data.external.identity_annadavi: Reading... data.external.identity_zoexue: Reading... data.external.identity_kechezon: Read complete after 0s [id=-] data.external.identity_jiyas: Reading... data.external.identity_qianxuege: Read complete after 0s [id=-] data.external.identity_kothurvidur: Reading... data.external.identity_jw8: Read complete after 0s [id=-] data.external.identity_thesuperrl: Reading... data.external.identity_acchs: Read complete after 0s [id=-] data.external.identity_nikhilr: Reading... data.external.identity_anish: Reading... data.external.identity_amzoeee: Reading... data.external.identity_stringatlas: Reading... data.external.identity_jessica_lr_song: Read complete after 1s [id=-] data.external.identity_krishsax: Reading... data.external.identity_zoexue: Read complete after 1s [id=-] data.external.identity_kothurvidur: Read complete after 1s [id=-] data.external.identity_etashjha: Reading... data.external.identity_arom1a: Reading... data.external.identity_jiyas: Read complete after 1s [id=-] data.external.identity_kluoxx: Reading... data.external.identity_annadavi: Read complete after 1s [id=-] data.external.identity_siddharthr: Reading... data.external.identity_thesuperrl: Read complete after 1s [id=-] data.external.identity_kritdass: Reading... data.external.identity_nikhilr: Read complete after 1s [id=-] data.external.identity_tentype: Reading... data.external.identity_anish: Read complete after 1s [id=-] data.external.identity_brightz: Reading... data.external.identity_amzoeee: Read complete after 1s [id=-] data.external.identity_rzhu3: Reading... data.external.identity_stringatlas: Read complete after 1s [id=-] data.external.identity_gabrielx: Reading... data.external.identity_krishsax: Read complete after 0s [id=-] data.external.identity_dkou: Reading... data.external.identity_kluoxx: Read complete after 0s [id=-] data.external.identity_jalenluorion: Reading... data.external.identity_arom1a: Read complete after 0s [id=-] data.external.identity_xboxbedrock: Reading... data.external.identity_etashjha: Read complete after 0s [id=-] data.external.identity_manrajs: Reading... data.external.identity_siddharthr: Read complete after 0s [id=-] data.external.identity_maybe_yiyi: Reading... data.external.identity_kritdass: Read complete after 0s [id=-] data.external.identity_jc_965: Reading... data.external.identity_tentype: Read complete after 0s [id=-] data.external.identity_dcrodrig: Reading... data.external.identity_brightz: Read complete after 0s [id=-] data.external.identity_emmawong: Reading... data.external.identity_rzhu3: Read complete after 0s [id=-] data.external.identity_riztofu: Reading... data.external.identity_gabrielx: Read complete after 0s [id=-] github_repository.quest_quest: Refreshing state... [id=quest] data.external.identity_xboxbedrock: Read complete after 0s [id=-] data.external.identity_dkou: Read complete after 0s [id=-] github_team.cmu-cal: Refreshing state... [id=15669112] github_repository.uia_scottylabs_org: Refreshing state... [id=scottylabs-org] data.external.identity_jalenluorion: Read complete after 0s [id=-] github_repository.cmu-cal_cal: Refreshing state... [id=cal] data.external.identity_manrajs: Read complete after 0s [id=-] data.external.identity_maybe_yiyi: Read complete after 0s [id=-] github_team.occ: Refreshing state... [id=19240310] github_team.terrier: Refreshing state... [id=12028577] data.external.identity_jc_965: Read complete after 0s [id=-] github_repository.slai_mcp_server: Refreshing state... [id=mcp-server] data.external.identity_dcrodrig: Read complete after 0s [id=-] github_team.cmu-eats: Refreshing state... [id=16807600] data.external.identity_emmawong: Read complete after 0s [id=-] github_repository.tartan-vote_tartan_vote: Refreshing state... [id=tartan-vote] data.external.identity_riztofu: Read complete after 0s [id=-] github_team.cmu-housing: Refreshing state... [id=17937919] github_repository.cmu-housing_housing: Refreshing state... [id=housing] github_team.lost-and-found: Refreshing state... [id=4088608] github_repository.cmu-eats_dining_api: Refreshing state... [id=dining-api] github_team.uia: Refreshing state... [id=17529763] github_repository.cmu-courses_courses: Refreshing state... [id=courses] github_repository.cbp_groupme_mirror: Refreshing state... [id=groupme-mirror] github_repository.cmu-courses_internet_archive: Refreshing state... [id=internet-archive] github_repository.slai_cmugpt_surface: Refreshing state... [id=cmugpt-surface] github_team.slai: Refreshing state... [id=17571328] github_repository.uia_components: Refreshing state... [id=components] github_team.cmu-insta: Refreshing state... [id=19240311] github_repository.devops_kennel: Refreshing state... [id=kennel] github_team.quest: Refreshing state... [id=17330197] github_repository.devops_documentation: Refreshing state... [id=documentation] github_repository.terrier_terrier: Refreshing state... [id=terrier] github_repository.slai_cmugpt_agent: Refreshing state... [id=cmugpt-agent] github_repository.devops_governance: Refreshing state... [id=governance] github_team.cbp: Refreshing state... [id=17170099] github_repository.cbp_dalmatian: Refreshing state... [id=dalmatian] github_repository.slai_cmugpt_sms_surface: Refreshing state... [id=cmugpt-sms-surface] github_repository.cmu-eats_eats: Refreshing state... [id=eats] github_team.cmu-courses: Refreshing state... [id=3643907] github_repository.devops_infrastructure: Refreshing state... [id=infrastructure] github_repository.devops_link_shortener: Refreshing state... [id=link-shortener] github_repository.devops_collie: Refreshing state... [id=collie] github_repository.cbp_bus_sign: Refreshing state... [id=bus-sign] github_repository.lost-and-found_lost_and_found_v2: Refreshing state... [id=lost-and-found-v2] github_repository.cmu-insta_insta: Refreshing state... [id=insta] github_repository.cbp_discord_verify: Refreshing state... [id=discord-verify] github_team.devops: Refreshing state... [id=14706218] github_team.tartan-vote: Refreshing state... [id=17701833] github_repository.occ_occ: Refreshing state... [id=occ] github_team.cmu-maps: Refreshing state... [id=10872360] github_repository.cmu-maps_maps: Refreshing state... [id=maps] forgejo_repository.cbp_bus_sign: Refreshing state... [name=bus-sign] forgejo_repository.slai_cmugpt_surface: Refreshing state... [name=cmugpt-surface] forgejo_repository.cbp_discord_verify: Refreshing state... [name=discord-verify] forgejo_repository.devops_collie: Refreshing state... [name=collie] forgejo_repository.uia_scottylabs_org: Refreshing state... [name=scottylabs-org] forgejo_repository.cmu-housing_housing: Refreshing state... [name=housing] forgejo_repository.cmu-courses_courses: Refreshing state... [name=courses] forgejo_repository.cbp_groupme_mirror: Refreshing state... [name=groupme-mirror] forgejo_repository.devops_kennel: Refreshing state... [name=kennel] forgejo_repository.slai_cmugpt_sms_surface: Refreshing state... [name=cmugpt-sms-surface] forgejo_repository.cmu-cal_cal: Refreshing state... [name=cal] forgejo_repository.uia_components: Refreshing state... [name=components] forgejo_repository.cbp_dalmatian: Refreshing state... [name=dalmatian] forgejo_repository.lost-and-found_lost_and_found_v2: Refreshing state... [name=lost-and-found-v2] forgejo_repository.cmu-maps_maps: Refreshing state... [name=maps] forgejo_repository.cmu-insta_insta: Refreshing state... [name=insta] forgejo_repository.tartan-vote_tartan_vote: Refreshing state... [name=tartan-vote] forgejo_repository.devops_governance: Refreshing state... [name=governance] forgejo_repository.cmu-eats_eats: Refreshing state... [name=eats] forgejo_repository.cmu-eats_dining_api: Refreshing state... [name=dining-api] forgejo_repository.occ_occ: Refreshing state... [name=occ] forgejo_repository.devops_documentation: Refreshing state... [name=documentation] forgejo_repository.terrier_terrier: Refreshing state... [name=terrier] forgejo_repository.quest_quest: Refreshing state... [name=quest] forgejo_repository.devops_link_shortener: Refreshing state... [name=link-shortener] forgejo_repository.slai_mcp_server: Refreshing state... [name=mcp-server] forgejo_repository.slai_cmugpt_agent: Refreshing state... [name=cmugpt-agent] data.forgejo_organization.this: Reading... forgejo_repository.devops_infrastructure: Refreshing state... [name=infrastructure] data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs] forgejo_repository.cmu-courses_internet_archive: Refreshing state... [name=internet-archive] github_team_membership.cmu-cal_jalenluorion: Refreshing state... [id=15669112:jalenluorion] github_team_membership.cmu-cal_anish: Refreshing state... [id=15669112:ap-1] github_team_membership.occ_xboxbedrock: Refreshing state... [id=19240310:xboxbedrock] github_team_membership.occ_dcrodrig: Refreshing state... [id=19240310:newhacker1746] github_team_membership.occ_acchs: Refreshing state... [id=19240310:acchs-at-cmu] github_team_membership.occ_anish: Refreshing state... [id=19240310:ap-1] github_team_membership.terrier_etashjha: Refreshing state... [id=12028577:etashj] github_team_membership.terrier_kothurvidur: Refreshing state... [id=12028577:kothurvidur] github_team_membership.terrier_dkou: Refreshing state... [id=12028577:sharkree] github_team_membership.terrier_amzoeee: Refreshing state... [id=12028577:amzoeee] github_team_membership.terrier_anish: Refreshing state... [id=12028577:ap-1] github_team_membership.terrier_kritdass: Refreshing state... [id=12028577:kritdass] github_team_membership.cmu-eats_arom1a: Refreshing state... [id=16807600:arom1a] github_team_membership.cmu-housing_nikhilr: Refreshing state... [id=17937919:ecstaticpilot] github_team_membership.cmu-housing_jessica_lr_song: Refreshing state... [id=17937919:jessica-lr-song] github_team_membership.cmu-housing_manrajs: Refreshing state... [id=17937919:manraj10] github_team_membership.cmu-housing_stringatlas: Refreshing state... [id=17937919:stringatlas] github_team_membership.cmu-housing_xboxbedrock: Refreshing state... [id=17937919:xboxbedrock] github_team_membership.cmu-housing_gostmeaper: Refreshing state... [id=17937919:gostmeaper] github_team_membership.cmu-housing_amzoeee: Refreshing state... [id=17937919:amzoeee] github_team_membership.cmu-housing_emmawong: Refreshing state... [id=17937919:emw868] github_team_membership.cmu-housing_rhsen: Refreshing state... [id=17937919:rohansen08] github_team_membership.cmu-housing_zoexue: Refreshing state... [id=17937919:o-zox-o] github_team_membership.cmu-housing_kluoxx: Refreshing state... [id=17937919:luokaden888] github_team_membership.cmu-housing_anish: Refreshing state... [id=17937919:ap-1] github_team_membership.cmu-housing_tentype: Refreshing state... [id=17937919:tentype] github_actions_repository_permissions.quest_quest_actions: Refreshing state... [id=quest] github_actions_repository_permissions.uia_scottylabs_org_actions: Refreshing state... [id=scottylabs-org] github_actions_repository_permissions.cmu-cal_cal_actions: Refreshing state... [id=cal] github_team_membership.lost-and-found_anish: Refreshing state... [id=4088608:ap-1] github_actions_repository_permissions.slai_mcp_server_actions: Refreshing state... [id=mcp-server] github_team_membership.uia_anish: Refreshing state... [id=17529763:ap-1] github_actions_repository_permissions.tartan-vote_tartan_vote_actions: Refreshing state... [id=tartan-vote] github_actions_repository_permissions.cmu-courses_courses_actions: Refreshing state... [id=courses] github_actions_repository_permissions.cbp_groupme_mirror_actions: Refreshing state... [id=groupme-mirror] github_actions_repository_permissions.cmu-housing_housing_actions: Refreshing state... [id=housing] github_actions_repository_permissions.cmu-courses_internet_archive_actions: Refreshing state... [id=internet-archive] github_actions_repository_permissions.slai_cmugpt_surface_actions: Refreshing state... [id=cmugpt-surface] github_team_membership.slai_krishsax: Refreshing state... [id=17571328:krishsaxena] github_team_membership.slai_xboxbedrock: Refreshing state... [id=17571328:xboxbedrock] github_team_membership.slai_qianxuege: Refreshing state... [id=17571328:qianxuege] github_team_membership.slai_rzhu3: Refreshing state... [id=17571328:richard-31415] github_team_membership.slai_jc_965: Refreshing state... [id=17571328:jc-965] github_team_membership.slai_anish: Refreshing state... [id=17571328:ap-1] github_team_membership.slai_jiyas: Refreshing state... [id=17571328:jiya2301] github_actions_repository_permissions.cmu-eats_dining_api_actions: Refreshing state... [id=dining-api] github_actions_repository_permissions.uia_components_actions: Refreshing state... [id=components] github_team_membership.cmu-insta_etashjha: Refreshing state... [id=19240311:etashj] github_team_membership.quest_kechezon: Refreshing state... [id=17330197:kechezon] github_team_membership.quest_brightz: Refreshing state... [id=17330197:bright-zheng] github_team_membership.quest_xboxbedrock: Refreshing state... [id=17330197:xboxbedrock] github_team_membership.quest_jw8: Refreshing state... [id=17330197:jeffreyyijiwang] github_team_membership.quest_anish: Refreshing state... [id=17330197:ap-1] github_actions_repository_permissions.slai_cmugpt_agent_actions: Refreshing state... [id=cmugpt-agent] github_team_membership.cbp_maybe_yiyi: Refreshing state... [id=17170099:maybe-yiyi] github_team_membership.cbp_liamwalk: Refreshing state... [id=17170099:pandaaured] github_team_membership.cbp_nikhilr: Refreshing state... [id=17170099:ecstaticpilot] github_team_membership.cbp_xboxbedrock: Refreshing state... [id=17170099:xboxbedrock] github_team_membership.cbp_anish: Refreshing state... [id=17170099:ap-1] github_team_membership.cbp_tentype: Refreshing state... [id=17170099:tentype] github_actions_repository_permissions.slai_cmugpt_sms_surface_actions: Refreshing state... [id=cmugpt-sms-surface] github_actions_repository_permissions.devops_kennel_actions: Refreshing state... [id=kennel] github_team_membership.cmu-courses_etashjha: Refreshing state... [id=3643907:etashj] github_team_membership.cmu-courses_dkou: Refreshing state... [id=3643907:sharkree] github_team_membership.cmu-courses_anish: Refreshing state... [id=3643907:ap-1] github_team_membership.cmu-courses_gabrielx: Refreshing state... [id=3643907:gabriel-xu] github_actions_repository_permissions.devops_documentation_actions: Refreshing state... [id=documentation] github_actions_repository_permissions.terrier_terrier_actions: Refreshing state... [id=terrier] github_actions_repository_permissions.devops_governance_actions: Refreshing state... [id=governance] github_actions_repository_permissions.cbp_dalmatian_actions: Refreshing state... [id=dalmatian] github_actions_repository_permissions.devops_collie_actions: Refreshing state... [id=collie] github_actions_repository_permissions.cmu-eats_eats_actions: Refreshing state... [id=eats] github_actions_repository_permissions.lost-and-found_lost_and_found_v2_actions: Refreshing state... [id=lost-and-found-v2] github_actions_repository_permissions.cbp_discord_verify_actions: Refreshing state... [id=discord-verify] restapi_object.cbp_bus_sign_kennel_webhook: Refreshing state... [id=21] restapi_object.cbp_bus_sign_push_mirror: Refreshing state... [id=remote_mirror_MkKxQ2vA9pc] restapi_object.slai_cmugpt_surface_push_mirror: Refreshing state... [id=remote_mirror_cPTQN6S5SpB] restapi_object.slai_cmugpt_surface_kennel_webhook: Refreshing state... [id=23] restapi_object.cbp_discord_verify_kennel_webhook: Refreshing state... [id=18] restapi_object.cbp_discord_verify_push_mirror: Refreshing state... [id=remote_mirror_3MV7Up6cY1X] restapi_object.devops_collie_push_mirror: Refreshing state... [id=remote_mirror_ljvVejNFXaS] restapi_object.devops_collie_kennel_webhook: Refreshing state... [id=17] restapi_object.uia_scottylabs_org_kennel_webhook: Refreshing state... [id=29] restapi_object.uia_scottylabs_org_push_mirror: Refreshing state... [id=remote_mirror_h6yN6QdS1lD] github_actions_repository_permissions.devops_infrastructure_actions: Refreshing state... [id=infrastructure] restapi_object.cmu-housing_housing_kennel_webhook: Refreshing state... [id=16] restapi_object.cmu-housing_housing_docs_webhook: Refreshing state... [id=15] restapi_object.cmu-housing_housing_push_mirror: Refreshing state... [id=remote_mirror_uoV2HG4aiOT] restapi_object.cmu-courses_courses_push_mirror: Refreshing state... [id=remote_mirror_WMR9yQwKFDF] restapi_object.cmu-courses_courses_kennel_webhook: Refreshing state... [id=11] restapi_object.cbp_groupme_mirror_push_mirror: Refreshing state... [id=remote_mirror_U61HJHB9VGE] restapi_object.devops_kennel_kennel_webhook: Refreshing state... [id=14] restapi_object.devops_kennel_push_mirror: Refreshing state... [id=remote_mirror_cT9psJkhTQ4] restapi_object.slai_cmugpt_sms_surface_kennel_webhook: Refreshing state... [id=19] restapi_object.slai_cmugpt_sms_surface_push_mirror: Refreshing state... [id=remote_mirror__fkBkASN2jA] restapi_object.cmu-cal_cal_kennel_webhook: Refreshing state... [id=20] restapi_object.cmu-cal_cal_push_mirror: Refreshing state... [id=remote_mirror_CO243uIONoq] restapi_object.uia_components_push_mirror: Refreshing state... [id=remote_mirror_Kwwa8TDnTgc] restapi_object.uia_components_kennel_webhook: Refreshing state... [id=27] restapi_object.cbp_dalmatian_kennel_webhook: Refreshing state... [id=6] restapi_object.cbp_dalmatian_docs_webhook: Refreshing state... [id=7] restapi_object.cbp_dalmatian_push_mirror: Refreshing state... [id=remote_mirror_qzlCpLK153F] restapi_object.lost-and-found_lost_and_found_v2_push_mirror: Refreshing state... [id=remote_mirror_n_UZBFU-CY4] restapi_object.cmu-maps_maps_push_mirror: Refreshing state... [id=remote_mirror_e4_FUnvoslJ] restapi_object.cmu-maps_maps_kennel_webhook: Refreshing state... [id=9] github_team_membership.devops_anish: Refreshing state... [id=14706218:ap-1] github_team_membership.devops_thesuperrl: Refreshing state... [id=14706218:thesuperrl] github_team_membership.devops_xboxbedrock: Refreshing state... [id=14706218:xboxbedrock] restapi_object.cmu-insta_insta_push_mirror: Refreshing state... [id=remote_mirror_891nu6XQ1us] restapi_object.tartan-vote_tartan_vote_kennel_webhook: Refreshing state... [id=24] restapi_object.tartan-vote_tartan_vote_push_mirror: Refreshing state... [id=remote_mirror_EUFq4sFSnH5] restapi_object.tartan-vote_tartan_vote_docs_webhook: Refreshing state... [id=25] restapi_object.devops_governance_push_mirror: Refreshing state... [id=remote_mirror_lt9JzkX1jMq] restapi_object.devops_governance_kennel_webhook: Refreshing state... [id=22] restapi_object.cmu-eats_eats_push_mirror: Refreshing state... [id=remote_mirror_sjYwMA6NPgh] restapi_object.cmu-eats_eats_kennel_webhook: Refreshing state... [id=31] restapi_object.cmu-eats_dining_api_push_mirror: Refreshing state... [id=remote_mirror_a4MsQ5rO1_Z] restapi_object.occ_occ_push_mirror: Refreshing state... [id=remote_mirror_moBLq1YG1qz] restapi_object.devops_documentation_push_mirror: Refreshing state... [id=remote_mirror_izmsqaNB6Wx] restapi_object.terrier_terrier_push_mirror: Refreshing state... [id=remote_mirror_YopDgZksZv5] restapi_object.terrier_terrier_kennel_webhook: Refreshing state... [id=12] github_actions_repository_permissions.devops_link_shortener_actions: Refreshing state... [id=link-shortener] restapi_object.quest_quest_push_mirror: Refreshing state... [id=remote_mirror_U9d9BPAon29] restapi_object.quest_quest_kennel_webhook: Refreshing state... [id=10] restapi_object.devops_link_shortener_push_mirror: Refreshing state... [id=remote_mirror_cp8daMpt82v] restapi_object.devops_link_shortener_kennel_webhook: Refreshing state... [id=8] restapi_object.slai_mcp_server_push_mirror: Refreshing state... [id=remote_mirror_Tqh2y0R_vDp] restapi_object.slai_mcp_server_kennel_webhook: Refreshing state... [id=13] forgejo_team.link-shortener_leads: Refreshing state... [name=link-shortener-leads] forgejo_team.cmugpt-surface_leads: Refreshing state... [name=cmugpt-surface-leads] forgejo_team.slai: Refreshing state... [name=slai] forgejo_team.maps_leads: Refreshing state... [name=maps-leads] forgejo_team.dalmatian_leads: Refreshing state... [name=dalmatian-leads] forgejo_team.cmugpt-sms-surface_leads: Refreshing state... [name=cmugpt-sms-surface-leads] forgejo_team.uia: Refreshing state... [name=uia] forgejo_team.housing_leads: Refreshing state... [name=housing-leads] forgejo_team.discord-verify_leads: Refreshing state... [name=discord-verify-leads] forgejo_team.cmu-housing: Refreshing state... [name=cmu-housing] forgejo_team.quest_leads: Refreshing state... [name=quest-leads] forgejo_team.documentation_leads: Refreshing state... [name=documentation-leads] forgejo_team.lost-and-found: Refreshing state... [name=lost-and-found] forgejo_team.terrier_leads: Refreshing state... [name=terrier-leads] forgejo_team.occ: Refreshing state... [name=occ] forgejo_team.occ-main_leads: Refreshing state... [name=occ-main-leads] forgejo_team.cbp: Refreshing state... [name=cbp] forgejo_team.tartan-vote_leads: Refreshing state... [name=tartan-vote-leads] forgejo_team.components_leads: Refreshing state... [name=components-leads] forgejo_team.cmu-courses: Refreshing state... [name=cmu-courses] forgejo_team.cmu-maps: Refreshing state... [name=cmu-maps] forgejo_team.scottylabs-org_leads: Refreshing state... [name=scottylabs-org-leads] forgejo_team.lost-and-found-v2_leads: Refreshing state... [name=lost-and-found-v2-leads] forgejo_team.bus-sign_leads: Refreshing state... [name=bus-sign-leads] forgejo_team.eats_leads: Refreshing state... [name=eats-leads] forgejo_team.terrier: Refreshing state... [name=terrier] forgejo_team.quest: Refreshing state... [name=quest] forgejo_team.devops: Refreshing state... [name=devops] forgejo_team.groupme-mirror_leads: Refreshing state... [name=groupme-mirror-leads] forgejo_team.tartan-vote: Refreshing state... [name=tartan-vote] forgejo_team.cmu-insta: Refreshing state... [name=cmu-insta] forgejo_team.cmugpt-agent_leads: Refreshing state... [name=cmugpt-agent-leads] forgejo_team.devops_leads: Refreshing state... [name=devops-leads] forgejo_team.cmugpt-mcp-server_leads: Refreshing state... [name=cmugpt-mcp-server-leads] forgejo_team.cmu-courses_leads: Refreshing state... [name=cmu-courses-leads] forgejo_team.insta_leads: Refreshing state... [name=insta-leads] forgejo_team.cmu-cal: Refreshing state... [name=cmu-cal] forgejo_team.cal_leads: Refreshing state... [name=cal-leads] forgejo_team.courses_leads: Refreshing state... [name=courses-leads] forgejo_team.cmu-eats: Refreshing state... [name=cmu-eats] restapi_object.slai_cmugpt_agent_push_mirror: Refreshing state... [id=remote_mirror_kNY6rVcxdyr] restapi_object.slai_cmugpt_agent_kennel_webhook: Refreshing state... [id=28] restapi_object.devops_infrastructure_push_mirror: Refreshing state... [id=remote_mirror_oYtFpKVzTaZ] restapi_object.devops_infrastructure_kennel_webhook: Refreshing state... [id=26] restapi_object.cmu-courses_internet_archive_push_mirror: Refreshing state... [id=remote_mirror_58X__Yfs0bR] github_team_membership.tartan-vote_anish: Refreshing state... [id=17701833:ap-1] github_team_membership.tartan-vote_maybe_yiyi: Refreshing state... [id=17701833:maybe-yiyi] github_team_membership.tartan-vote_riztofu: Refreshing state... [id=17701833:riztofu] github_team_membership.tartan-vote_tentype: Refreshing state... [id=17701833:tentype] github_team_membership.tartan-vote_xboxbedrock: Refreshing state... [id=17701833:xboxbedrock] github_team_membership.cmu-maps_siddharthr: Refreshing state... [id=10872360:overthrowing] github_team_membership.cmu-maps_anish: Refreshing state... [id=10872360:ap-1] github_actions_repository_permissions.cbp_bus_sign_actions: Refreshing state... [id=bus-sign] github_actions_repository_permissions.cmu-insta_insta_actions: Refreshing state... [id=insta] github_actions_repository_permissions.occ_occ_actions: Refreshing state... [id=occ] github_actions_repository_permissions.cmu-maps_maps_actions: Refreshing state... [id=maps] github_repository_deploy_key.cbp_bus_sign_mirror_deploy_key: Refreshing state... [id=bus-sign:159665342] github_repository_deploy_key.slai_cmugpt_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-surface:159665348] github_repository_deploy_key.cbp_discord_verify_mirror_deploy_key: Refreshing state... [id=discord-verify:159665340] github_repository_deploy_key.devops_collie_mirror_deploy_key: Refreshing state... [id=collie:159665332] github_repository_deploy_key.uia_scottylabs_org_mirror_deploy_key: Refreshing state... [id=scottylabs-org:159665361] github_repository_deploy_key.cmu-housing_housing_mirror_deploy_key: Refreshing state... [id=housing:159665369] github_repository_deploy_key.cmu-courses_courses_mirror_deploy_key: Refreshing state... [id=courses:159665327] github_repository_deploy_key.cbp_groupme_mirror_mirror_deploy_key: Refreshing state... [id=groupme-mirror:159665366] github_repository_deploy_key.devops_kennel_mirror_deploy_key: Refreshing state... [id=kennel:159665328] github_repository_deploy_key.slai_cmugpt_sms_surface_mirror_deploy_key: Refreshing state... [id=cmugpt-sms-surface:159665365] github_repository_deploy_key.cmu-cal_cal_mirror_deploy_key: Refreshing state... [id=cal:159665344] github_repository_deploy_key.uia_components_mirror_deploy_key: Refreshing state... [id=components:159665353] github_repository_deploy_key.cbp_dalmatian_mirror_deploy_key: Refreshing state... [id=dalmatian:159665367] github_repository_deploy_key.lost-and-found_lost_and_found_v2_mirror_deploy_key: Refreshing state... [id=lost-and-found-v2:159665337] github_repository_deploy_key.cmu-maps_maps_mirror_deploy_key: Refreshing state... [id=maps:159665325] github_repository_deploy_key.cmu-insta_insta_mirror_deploy_key: Refreshing state... [id=insta:161720245] github_repository_deploy_key.tartan-vote_tartan_vote_mirror_deploy_key: Refreshing state... [id=tartan-vote:159665350] github_repository_deploy_key.devops_governance_mirror_deploy_key: Refreshing state... [id=governance:159665346] github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Refreshing state... [id=eats:161775995] github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Refreshing state... [id=dining-api:161772815] github_repository_deploy_key.occ_occ_mirror_deploy_key: Refreshing state... [id=occ:161720242] github_repository_deploy_key.devops_documentation_mirror_deploy_key: Refreshing state... [id=documentation:159665362] github_repository_deploy_key.terrier_terrier_mirror_deploy_key: Refreshing state... [id=terrier:159665339] github_repository_deploy_key.quest_quest_mirror_deploy_key: Refreshing state... [id=quest:159665323] github_repository_deploy_key.devops_link_shortener_mirror_deploy_key: Refreshing state... [id=link-shortener:159665345] github_repository_deploy_key.slai_mcp_server_mirror_deploy_key: Refreshing state... [id=mcp-server:159665358] forgejo_team_member.cmugpt-surface_leads_anish: Refreshing state... forgejo_team_member.cmugpt-surface_leads_krishsax: Refreshing state... forgejo_team_member.link-shortener_leads_anish: Refreshing state... forgejo_team_member.link-shortener_leads_thesuperrl: Refreshing state... forgejo_team_repository.slai_cmugpt_agent: Refreshing state... forgejo_team_member.slai_jiyas: Refreshing state... forgejo_team_repository.slai_cmugpt_surface: Refreshing state... forgejo_team_member.slai_rzhu3: Refreshing state... forgejo_team_repository.slai_cmugpt_sms_surface: Refreshing state... forgejo_team_member.slai_qianxuege: Refreshing state... forgejo_team_repository.slai_mcp_server: Refreshing state... forgejo_team_member.slai_anish: Refreshing state... forgejo_team_member.slai_jc_965: Refreshing state... forgejo_team_member.slai_xboxbedrock: Refreshing state... forgejo_team_member.slai_krishsax: Refreshing state... forgejo_team_member.maps_leads_siddharthr: Refreshing state... forgejo_team_member.maps_leads_anish: Refreshing state... forgejo_team_member.dalmatian_leads_anish: Refreshing state... forgejo_team_member.dalmatian_leads_maybe_yiyi: Refreshing state... forgejo_team_member.dalmatian_leads_tentype: Refreshing state... forgejo_team_member.cmugpt-sms-surface_leads_krishsax: Refreshing state... forgejo_team_member.cmugpt-sms-surface_leads_anish: Refreshing state... forgejo_team_member.uia_anish: Refreshing state... forgejo_team_repository.uia_scottylabs_org: Refreshing state... forgejo_team_repository.uia_components: Refreshing state... forgejo_team_member.housing_leads_anish: Refreshing state... forgejo_team_member.housing_leads_gostmeaper: Refreshing state... forgejo_team_member.housing_leads_nikhilr: Refreshing state... forgejo_team_member.housing_leads_tentype: Refreshing state... forgejo_team_member.housing_leads_emmawong: Refreshing state... forgejo_team_member.discord-verify_leads_anish: Refreshing state... forgejo_team_member.discord-verify_leads_maybe_yiyi: Refreshing state... forgejo_team_member.discord-verify_leads_tentype: Refreshing state... forgejo_team_member.cmu-housing_stringatlas: Refreshing state... forgejo_team_member.cmu-housing_zoexue: Refreshing state... forgejo_team_member.cmu-housing_rhsen: Refreshing state... forgejo_team_member.cmu-housing_gostmeaper: Refreshing state... forgejo_team_member.cmu-housing_xboxbedrock: Refreshing state... forgejo_team_repository.cmu-housing_housing: Refreshing state... forgejo_team_member.cmu-housing_tentype: Refreshing state... forgejo_team_member.cmu-housing_emmawong: Refreshing state... forgejo_team_member.cmu-housing_kluoxx: Refreshing state... forgejo_team_member.cmu-housing_amzoeee: Refreshing state... forgejo_team_member.cmu-housing_nikhilr: Refreshing state... forgejo_team_member.cmu-housing_anish: Refreshing state... forgejo_team_member.cmu-housing_manrajs: Refreshing state... forgejo_team_member.cmu-housing_jessica_lr_song: Refreshing state... forgejo_team_member.quest_leads_anish: Refreshing state... forgejo_team_member.quest_leads_xboxbedrock: Refreshing state... forgejo_team_member.documentation_leads_thesuperrl: Refreshing state... forgejo_team_member.documentation_leads_anish: Refreshing state... forgejo_team_member.lost-and-found_anish: Refreshing state... forgejo_team_repository.lost-and-found_lost_and_found_v2: Refreshing state... forgejo_team_member.terrier_leads_anish: Refreshing state... forgejo_team_member.occ_xboxbedrock: Refreshing state... forgejo_team_member.terrier_leads_kritdass: Refreshing state... forgejo_team_member.occ_acchs: Refreshing state... forgejo_team_repository.occ_occ: Refreshing state... forgejo_team_member.occ_dcrodrig: Refreshing state... forgejo_team_member.occ_anish: Refreshing state... forgejo_team_member.occ-main_leads_dcrodrig: Refreshing state... forgejo_team_member.occ-main_leads_anish: Refreshing state... forgejo_team_repository.cbp_bus_sign: Refreshing state... forgejo_team_repository.cbp_discord_verify: Refreshing state... forgejo_team_repository.cbp_dalmatian: Refreshing state... forgejo_team_member.cbp_maybe_yiyi: Refreshing state... forgejo_team_member.cbp_tentype: Refreshing state... forgejo_team_member.cbp_anish: Refreshing state... forgejo_team_member.cbp_liamwalk: Refreshing state... forgejo_team_member.cbp_xboxbedrock: Refreshing state... forgejo_team_repository.cbp_groupme_mirror: Refreshing state... forgejo_team_member.cbp_nikhilr: Refreshing state... forgejo_team_member.tartan-vote_leads_maybe_yiyi: Refreshing state... forgejo_team_member.tartan-vote_leads_anish: Refreshing state... forgejo_team_member.components_leads_anish: Refreshing state... forgejo_team_member.cmu-courses_etashjha: Refreshing state... forgejo_team_member.cmu-courses_dkou: Refreshing state... forgejo_team_repository.cmu-courses_courses: Refreshing state... forgejo_team_member.cmu-courses_gabrielx: Refreshing state... forgejo_team_repository.cmu-courses_internet_archive: Refreshing state... forgejo_team_member.cmu-courses_anish: Refreshing state... forgejo_team_repository.cmu-maps_maps: Refreshing state... forgejo_team_member.cmu-maps_anish: Refreshing state... forgejo_team_member.cmu-maps_siddharthr: Refreshing state... forgejo_team_member.scottylabs-org_leads_anish: Refreshing state... forgejo_team_member.lost-and-found-v2_leads_anish: Refreshing state... forgejo_team_member.bus-sign_leads_anish: Refreshing state... forgejo_team_member.bus-sign_leads_tentype: Refreshing state... forgejo_team_member.bus-sign_leads_maybe_yiyi: Refreshing state... forgejo_team_member.terrier_kritdass: Refreshing state... forgejo_team_member.terrier_dkou: Refreshing state... forgejo_team_member.terrier_kothurvidur: Refreshing state... forgejo_team_member.terrier_amzoeee: Refreshing state... forgejo_team_member.terrier_etashjha: Refreshing state... forgejo_team_member.terrier_anish: Refreshing state... forgejo_team_repository.terrier_terrier: Refreshing state... forgejo_team_member.eats_leads_arom1a: Refreshing state... forgejo_team_member.quest_jw8: Refreshing state... forgejo_team_member.quest_kechezon: Refreshing state... forgejo_team_member.quest_brightz: Refreshing state... forgejo_team_member.quest_anish: Refreshing state... forgejo_team_member.quest_xboxbedrock: Refreshing state... forgejo_team_repository.quest_quest: Refreshing state... forgejo_team_repository.devops_kennel: Refreshing state... forgejo_team_repository.devops_collie: Refreshing state... forgejo_team_repository.devops_infrastructure: Refreshing state... forgejo_team_member.devops_anish: Refreshing state... forgejo_team_repository.devops_documentation: Refreshing state... forgejo_team_member.devops_thesuperrl: Refreshing state... forgejo_team_repository.devops_governance: Refreshing state... forgejo_team_member.devops_xboxbedrock: Refreshing state... forgejo_team_repository.devops_link_shortener: Refreshing state... forgejo_team_member.groupme-mirror_leads_maybe_yiyi: Refreshing state... forgejo_team_member.groupme-mirror_leads_tentype: Refreshing state... forgejo_team_member.groupme-mirror_leads_anish: Refreshing state... forgejo_team_member.tartan-vote_xboxbedrock: Refreshing state... forgejo_team_repository.tartan-vote_tartan_vote: Refreshing state... forgejo_team_member.tartan-vote_riztofu: Refreshing state... forgejo_team_member.tartan-vote_anish: Refreshing state... forgejo_team_member.tartan-vote_tentype: Refreshing state... forgejo_team_member.tartan-vote_maybe_yiyi: Refreshing state... forgejo_team_member.cmugpt-agent_leads_anish: Refreshing state... forgejo_team_member.cmugpt-agent_leads_krishsax: Refreshing state... forgejo_team_repository.cmu-insta_insta: Refreshing state... forgejo_team_member.cmu-insta_etashjha: Refreshing state... forgejo_team_member.devops_leads_thesuperrl: Refreshing state... forgejo_team_member.devops_leads_anish: Refreshing state... forgejo_team_member.cmugpt-mcp-server_leads_krishsax: Refreshing state... forgejo_team_member.cmugpt-mcp-server_leads_anish: Refreshing state... forgejo_team_member.cmu-courses_leads_anish: Refreshing state... forgejo_team_member.cmu-courses_leads_etashjha: Refreshing state... forgejo_team_member.insta_leads_etashjha: Refreshing state... forgejo_team_member.cmu-cal_jalenluorion: Refreshing state... forgejo_team_member.cmu-cal_anish: Refreshing state... forgejo_team_member.cal_leads_jalenluorion: Refreshing state... forgejo_team_repository.cmu-cal_cal: Refreshing state... forgejo_team_member.cal_leads_anish: Refreshing state... forgejo_team_member.courses_leads_anish: Refreshing state... forgejo_team_member.cmu-eats_arom1a: Refreshing state... forgejo_team_member.courses_leads_etashjha: Refreshing state... github_repository_deploy_key.slai_cmugpt_agent_mirror_deploy_key: Refreshing state... [id=cmugpt-agent:159665360] github_repository_deploy_key.devops_infrastructure_mirror_deploy_key: Refreshing state... [id=infrastructure:159665343] github_repository_deploy_key.cmu-courses_internet_archive_mirror_deploy_key: Refreshing state... [id=internet-archive:159665356] github_repository_ruleset.cbp_bus_sign_ruleset: Refreshing state... [id=15236550] github_repository_ruleset.slai_cmugpt_surface_ruleset: Refreshing state... [id=16476662] github_repository_ruleset.cbp_discord_verify_ruleset: Refreshing state... [id=11449154] github_repository_ruleset.devops_collie_ruleset: Refreshing state... [id=18865204] github_repository_ruleset.uia_scottylabs_org_ruleset: Refreshing state... [id=18112214] github_repository_ruleset.cmu-housing_housing_ruleset: Refreshing state... [id=17421604] github_repository_ruleset.cmu-courses_courses_ruleset: Refreshing state... [id=15855161] github_repository_ruleset.cbp_groupme_mirror_ruleset: Refreshing state... [id=12503296] github_repository_ruleset.devops_kennel_ruleset: Refreshing state... [id=13193453] github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Refreshing state... [id=17653622] github_repository_ruleset.cmu-cal_cal_ruleset: Refreshing state... [id=18919281] github_repository_ruleset.uia_components_ruleset: Refreshing state... [id=16366085] github_repository_ruleset.cbp_dalmatian_ruleset: Refreshing state... [id=17238915] github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Refreshing state... [id=18920618] github_repository_ruleset.cmu-maps_maps_ruleset: Refreshing state... [id=18864961] github_repository_ruleset.cmu-insta_insta_ruleset: Refreshing state... [id=21841636] github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Refreshing state... [id=16848966] forgejo_team_repository.cmugpt-agent_leads_cmugpt_agent: Refreshing state... forgejo_team_repository.cmugpt-surface_leads_cmugpt_surface: Refreshing state... forgejo_team_repository.cmugpt-sms-surface_leads_cmugpt_sms_surface: Refreshing state... forgejo_team_repository.cmugpt-mcp-server_leads_mcp_server: Refreshing state... github_repository_ruleset.devops_governance_ruleset: Refreshing state... [id=14488792] forgejo_team_repository.scottylabs-org_leads_scottylabs_org: Refreshing state... forgejo_team_repository.components_leads_components: Refreshing state... github_repository_ruleset.cmu-eats_eats_ruleset: Refreshing state... [id=21330218] forgejo_team_repository.housing_leads_housing: Refreshing state... forgejo_team_repository.lost-and-found-v2_leads_lost_and_found_v2: Refreshing state... github_repository_ruleset.cmu-eats_dining_api_ruleset: Refreshing state... [id=21879626] forgejo_team_repository.occ-main_leads_occ: Refreshing state... forgejo_team_repository.bus-sign_leads_bus_sign: Refreshing state... forgejo_team_repository.discord-verify_leads_discord_verify: Refreshing state... forgejo_team_repository.dalmatian_leads_dalmatian: Refreshing state... forgejo_team_repository.groupme-mirror_leads_groupme_mirror: Refreshing state... github_repository_ruleset.occ_occ_ruleset: Refreshing state... [id=21841634] forgejo_team_repository.courses_leads_courses: Refreshing state... forgejo_team_repository.cmu-courses_leads_internet_archive: Refreshing state... forgejo_team_repository.maps_leads_maps: Refreshing state... forgejo_team_repository.terrier_leads_terrier: Refreshing state... github_repository_ruleset.devops_documentation_ruleset: Refreshing state... [id=17079368] forgejo_team_repository.quest_leads_quest: Refreshing state... forgejo_team_repository.devops_leads_kennel: Refreshing state... forgejo_team_repository.documentation_leads_documentation: Refreshing state... forgejo_team_repository.devops_leads_collie: Refreshing state... forgejo_team_repository.devops_leads_infrastructure: Refreshing state... forgejo_team_repository.devops_leads_governance: Refreshing state... forgejo_team_repository.link-shortener_leads_link_shortener: Refreshing state... forgejo_team_repository.tartan-vote_leads_tartan_vote: Refreshing state... forgejo_team_repository.insta_leads_insta: Refreshing state... github_repository_ruleset.terrier_terrier_ruleset: Refreshing state... [id=12671003] forgejo_team_repository.eats_leads_eats: Refreshing state... forgejo_team_repository.eats_leads_dining_api: Refreshing state... forgejo_team_repository.cal_leads_cal: Refreshing state... github_repository_ruleset.quest_quest_ruleset: Refreshing state... [id=15750052] github_repository_ruleset.devops_link_shortener_ruleset: Refreshing state... [id=17911647] github_repository_ruleset.slai_mcp_server_ruleset: Refreshing state... [id=17082154] github_repository_ruleset.slai_cmugpt_agent_ruleset: Refreshing state... [id=16476661] github_repository_ruleset.devops_infrastructure_ruleset: Refreshing state... [id=11275118] github_repository_ruleset.cmu-courses_internet_archive_ruleset: Refreshing state... [id=11988074] Note: Objects have changed outside of OpenTofu OpenTofu detected the following changes made outside of OpenTofu since the last "tofu apply" which may have affected this plan: # restapi_object.cmu-eats_dining_api_push_mirror has been deleted - resource "restapi_object" "cmu-eats_dining_api_push_mirror" { - api_data = { - "public_key" = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIIKyjdH3wH8J3aSgeZMLB9/uUGzyYltQsk5Cl1Q8MXqF" # (9 unchanged elements hidden) } -> null id = "remote_mirror_a4MsQ5rO1_Z" # (5 unchanged attributes hidden) } # restapi_object.cmu-eats_eats_push_mirror has been deleted - resource "restapi_object" "cmu-eats_eats_push_mirror" { - api_data = { - "public_key" = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFsm79Zlyi0JKH4y/BdgCCoLyrC4/18yF3owfsqOt53/" # (9 unchanged elements hidden) } -> null id = "remote_mirror_sjYwMA6NPgh" # (5 unchanged attributes hidden) } Unless you have made equivalent changes to your configuration, or ignored the relevant attributes using ignore_changes, the following plan may include actions to undo or respond to these changes. ───────────────────────────────────────────────────────────────────────────── OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ~ update in-place (current -> planned) -/+ destroy and then create replacement OpenTofu planned the following actions, but then encountered a problem: # forgejo_repository.cmu-eats_dining_api will be updated in-place ~ resource "forgejo_repository" "cmu-eats_dining_api" { ~ clone_url = "https://git.cmu.dev/ScottyLabs/dining-api.git" -> (known after apply) ~ created_at = "2026-09-18T23:16:27Z" -> (known after apply) + description = "Dining API for CMU dining locations and availability" ~ empty = false -> (known after apply) + external_tracker = (known after apply) + external_wiki = (known after apply) ~ forks_count = 0 -> (known after apply) ~ full_name = "ScottyLabs/dining-api" -> (known after apply) ~ html_url = "https://git.cmu.dev/ScottyLabs/dining-api" -> (known after apply) id = 97 ~ internal_tracker = { ~ allow_only_contributors_to_track_time = true -> (known after apply) ~ enable_issue_dependencies = true -> (known after apply) ~ enable_time_tracker = true -> (known after apply) } -> (known after apply) + mirror_interval = (known after apply) ~ mirror_updated = "0001-01-01T00:00:00Z" -> (known after apply) name = "dining-api" ~ open_issues_count = 0 -> (known after apply) ~ open_pr_counter = 5 -> (known after apply) ~ permissions = { ~ admin = true -> (known after apply) ~ pull = true -> (known after apply) ~ push = true -> (known after apply) } -> (known after apply) ~ release_counter = 0 -> (known after apply) ~ size = 38816 -> (known after apply) ~ ssh_url = "ssh://forgejo@git.cmu.dev/ScottyLabs/dining-api.git" -> (known after apply) ~ stars_count = 0 -> (known after apply) ~ updated_at = "2026-09-18T23:22:22Z" -> (known after apply) ~ watchers_count = 3 -> (known after apply) ~ website = "https://git.cmu.dev/ScottyLabs/dining-api" -> "https://api.cmueats.com" # (30 unchanged attributes hidden) } # forgejo_repository.cmu-eats_eats will be updated in-place ~ resource "forgejo_repository" "cmu-eats_eats" { ~ clone_url = "https://git.cmu.dev/ScottyLabs/eats.git" -> (known after apply) ~ created_at = "2026-09-18T23:15:39Z" -> (known after apply) ~ description = "Vite web app for CMU dining location info" -> "CMU dining location info" ~ empty = false -> (known after apply) + external_tracker = (known after apply) + external_wiki = (known after apply) ~ forks_count = 0 -> (known after apply) ~ full_name = "ScottyLabs/eats" -> (known after apply) ~ html_url = "https://git.cmu.dev/ScottyLabs/eats" -> (known after apply) id = 96 ~ internal_tracker = { ~ allow_only_contributors_to_track_time = true -> (known after apply) ~ enable_issue_dependencies = true -> (known after apply) ~ enable_time_tracker = true -> (known after apply) } -> (known after apply) + mirror_interval = (known after apply) ~ mirror_updated = "0001-01-01T00:00:00Z" -> (known after apply) name = "eats" ~ open_issues_count = 48 -> (known after apply) ~ open_pr_counter = 4 -> (known after apply) ~ permissions = { ~ admin = true -> (known after apply) ~ pull = true -> (known after apply) ~ push = true -> (known after apply) } -> (known after apply) ~ release_counter = 0 -> (known after apply) ~ size = 95496 -> (known after apply) ~ ssh_url = "ssh://forgejo@git.cmu.dev/ScottyLabs/eats.git" -> (known after apply) ~ stars_count = 0 -> (known after apply) ~ updated_at = "2026-09-18T23:35:12Z" -> (known after apply) ~ watchers_count = 3 -> (known after apply) # (31 unchanged attributes hidden) } # forgejo_repository.cmu-study_study will be created + resource "forgejo_repository" "cmu-study_study" { + allow_manual_merge = false + allow_merge_commits = true + allow_rebase = true + allow_rebase_explicit = true + allow_squash_merge = true + archive_on_destroy = false + archived = false + auto_init = true + autodetect_manual_merge = false + avatar_url = (known after apply) + clone_url = (known after apply) + created_at = (known after apply) + default_branch = "main" + default_merge_style = "merge" + description = "A platform for creating and joining study groups" + empty = (known after apply) + external_tracker = (known after apply) + external_wiki = (known after apply) + fork = (known after apply) + forks_count = (known after apply) + full_name = (known after apply) + has_actions = true + has_issues = true + has_packages = true + has_projects = true + has_pull_requests = true + has_releases = true + has_wiki = true + html_url = (known after apply) + id = (known after apply) + ignore_whitespace_conflicts = false + internal = (known after apply) + internal_tracker = (known after apply) + labels = false + lfs = false + milestones = false + mirror = false + mirror_interval = (known after apply) + mirror_updated = (known after apply) + name = "study" + open_issues_count = (known after apply) + open_pr_counter = (known after apply) + owner = "ScottyLabs" + parent_id = (known after apply) + permissions = (known after apply) + private = false + release_counter = (known after apply) + size = (known after apply) + ssh_url = (known after apply) + stars_count = (known after apply) + template = false + trust_model = "default" + updated_at = (known after apply) + watchers_count = (known after apply) + website = "https://cmustudy.com" } # forgejo_team.cmu-study will be created + resource "forgejo_team" "cmu-study" { + can_create_org_repo = (known after apply) + description = "CMU Study" + id = (known after apply) + includes_all_repositories = (known after apply) + name = "cmu-study" + organization = (known after apply) + organization_id = 3 + permission = "write" + units_map = { + "repo.code" = "write" + "repo.issues" = "write" + "repo.pulls" = "write" } } # forgejo_team.study_leads will be created + resource "forgejo_team" "study_leads" { + can_create_org_repo = false + description = "CMU Study leads" + id = (known after apply) + includes_all_repositories = false + name = "study-leads" + organization = (known after apply) + organization_id = 3 + permission = "admin" + units_map = { + "repo.actions" = "admin" + "repo.code" = "admin" + "repo.ext_issues" = "admin" + "repo.ext_wiki" = "admin" + "repo.issues" = "admin" + "repo.packages" = "admin" + "repo.projects" = "admin" + "repo.pulls" = "admin" + "repo.releases" = "admin" + "repo.wiki" = "admin" } } # forgejo_team_member.cmu-study_annadavi will be created + resource "forgejo_team_member" "cmu-study_annadavi" { + team_id = (known after apply) + user = "annadavi" } # forgejo_team_member.study_leads_annadavi will be created + resource "forgejo_team_member" "study_leads_annadavi" { + team_id = (known after apply) + user = "annadavi" } # forgejo_team_repository.cmu-eats_dining_api will be created + resource "forgejo_team_repository" "cmu-eats_dining_api" { + owner = "ScottyLabs" + repository = "dining-api" + team_id = 28 } # forgejo_team_repository.cmu-eats_eats will be created + resource "forgejo_team_repository" "cmu-eats_eats" { + owner = "ScottyLabs" + repository = "eats" + team_id = 28 } # forgejo_team_repository.cmu-study_study will be created + resource "forgejo_team_repository" "cmu-study_study" { + owner = "ScottyLabs" + repository = "study" + team_id = (known after apply) } # forgejo_team_repository.study_leads_study will be created + resource "forgejo_team_repository" "study_leads_study" { + owner = "ScottyLabs" + repository = "study" + team_id = (known after apply) } # github_actions_repository_permissions.cmu-study_study_actions will be created + resource "github_actions_repository_permissions" "cmu-study_study_actions" { + enabled = false + id = (known after apply) + repository = "study" + sha_pinning_required = (known after apply) } # github_repository.cmu-study_study will be created + resource "github_repository" "cmu-study_study" { + allow_auto_merge = false + allow_forking = (known after apply) + allow_merge_commit = true + allow_rebase_merge = true + allow_squash_merge = true + archived = false + default_branch = (known after apply) + delete_branch_on_merge = false + description = "READ-ONLY MIRROR" + etag = (known after apply) + fork = (known after apply) + full_name = (known after apply) + git_clone_url = (known after apply) + has_discussions = false + has_issues = false + has_projects = false + has_wiki = false + homepage_url = "https://git.cmu.dev/ScottyLabs/study" + html_url = (known after apply) + http_clone_url = (known after apply) + id = (known after apply) + ignore_vulnerability_alerts_during_read = false + merge_commit_message = "PR_TITLE" + merge_commit_title = "MERGE_MESSAGE" + name = "study" + node_id = (known after apply) + primary_language = (known after apply) + private = (known after apply) + repo_id = (known after apply) + source_owner = (known after apply) + source_repo = (known after apply) + squash_merge_commit_message = "COMMIT_MESSAGES" + squash_merge_commit_title = "COMMIT_OR_PR_TITLE" + ssh_clone_url = (known after apply) + svn_url = (known after apply) + topics = (known after apply) + visibility = "public" + vulnerability_alerts = (known after apply) + web_commit_signoff_required = (known after apply) + security_and_analysis (known after apply) } # github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key must be replaced -/+ resource "github_repository_deploy_key" "cmu-eats_dining_api_mirror_deploy_key" { ~ etag = "W/\"bf394a0fcc9e1231b4ee544598e6680359d71d177764d9f0c8cd22f5c7ec3ef4\"" -> (known after apply) ~ id = "dining-api:161772815" -> (known after apply) ~ key = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIIKyjdH3wH8J3aSgeZMLB9/uUGzyYltQsk5Cl1Q8MXqF" # forces replacement -> (known after apply) # forces replacement # (3 unchanged attributes hidden) } # github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key must be replaced -/+ resource "github_repository_deploy_key" "cmu-eats_eats_mirror_deploy_key" { ~ etag = "W/\"5cf445e3b52f1b2e59fe1cb32c5c2e0859e2598566a7aa75c3ebdbf4b01f6164\"" -> (known after apply) ~ id = "eats:161775995" -> (known after apply) ~ key = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFsm79Zlyi0JKH4y/BdgCCoLyrC4/18yF3owfsqOt53/" # forces replacement -> (known after apply) # forces replacement # (3 unchanged attributes hidden) } # github_repository_deploy_key.cmu-study_study_mirror_deploy_key will be created + resource "github_repository_deploy_key" "cmu-study_study_mirror_deploy_key" { + etag = (known after apply) + id = (known after apply) + key = (known after apply) + read_only = false + repository = "study" + title = "Forgejo mirror" } # github_repository_ruleset.cmu-eats_dining_api_ruleset will be updated in-place ~ resource "github_repository_ruleset" "cmu-eats_dining_api_ruleset" { id = "21879626" name = "Default" # (6 unchanged attributes hidden) ~ bypass_actors { ~ actor_id = 161772815 -> (known after apply) # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place ~ resource "github_repository_ruleset" "cmu-eats_eats_ruleset" { id = "21330218" name = "Default" # (6 unchanged attributes hidden) ~ bypass_actors { ~ actor_id = 161775995 -> (known after apply) # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.cmu-study_study_ruleset will be created + resource "github_repository_ruleset" "cmu-study_study_ruleset" { + enforcement = "active" + etag = (known after apply) + id = (known after apply) + name = "Default" + node_id = (known after apply) + repository = "study" + ruleset_id = (known after apply) + target = "branch" + bypass_actors { + actor_id = (known after apply) + actor_type = "DeployKey" + bypass_mode = "always" } + conditions { + ref_name { + exclude = [] + include = [ + "~ALL", ] } } + rules { + creation = true + deletion = true + non_fast_forward = true + update = true + update_allows_fetch_and_merge = false + pull_request { + allowed_merge_methods = (known after apply) + dismiss_stale_reviews_on_push = false + require_code_owner_review = false + require_last_push_approval = false + required_approving_review_count = 1 + required_review_thread_resolution = false } } } # github_team.cmu-study will be created + resource "github_team" "cmu-study" { + create_default_maintainer = false + etag = (known after apply) + id = (known after apply) + members_count = (known after apply) + name = "CMU Study" + node_id = (known after apply) + notification_setting = "notifications_enabled" + parent_team_read_id = (known after apply) + parent_team_read_slug = (known after apply) + privacy = "closed" + slug = (known after apply) } # github_team_membership.cmu-study_annadavi will be created + resource "github_team_membership" "cmu-study_annadavi" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = (known after apply) + username = "annadavid1" } # restapi_object.cmu-eats_dining_api_push_mirror will be created + resource "restapi_object" "cmu-eats_dining_api_push_mirror" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = jsonencode( { + interval = "8h0m0s" + remote_address = "git@github.com:ScottyLabs/dining-api.git" + sync_on_commit = true + use_ssh = true } ) + id = (known after apply) + id_attribute = "remote_name" + path = "/api/v1/repos/ScottyLabs/dining-api/push_mirrors" } # restapi_object.cmu-eats_eats_kennel_webhook will be created + resource "restapi_object" "cmu-eats_eats_kennel_webhook" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = (sensitive value) + id = (known after apply) + id_attribute = "id" + path = "/api/v1/repos/ScottyLabs/eats/hooks" + update_method = "PATCH" } # restapi_object.cmu-eats_eats_push_mirror will be created + resource "restapi_object" "cmu-eats_eats_push_mirror" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = jsonencode( { + interval = "8h0m0s" + remote_address = "git@github.com:ScottyLabs/eats.git" + sync_on_commit = true + use_ssh = true } ) + id = (known after apply) + id_attribute = "remote_name" + path = "/api/v1/repos/ScottyLabs/eats/push_mirrors" } # restapi_object.cmu-study_study_kennel_webhook will be created + resource "restapi_object" "cmu-study_study_kennel_webhook" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = (sensitive value) + id = (known after apply) + id_attribute = "id" + path = "/api/v1/repos/ScottyLabs/study/hooks" + update_method = "PATCH" } # restapi_object.cmu-study_study_push_mirror will be created + resource "restapi_object" "cmu-study_study_push_mirror" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = jsonencode( { + interval = "8h0m0s" + remote_address = "git@github.com:ScottyLabs/study.git" + sync_on_commit = true + use_ssh = true } ) + id = (known after apply) + id_attribute = "remote_name" + path = "/api/v1/repos/ScottyLabs/study/push_mirrors" } Plan: 22 to add, 4 to change, 2 to destroy. ╷ │ Error: Unable to read team repository │ │ with forgejo_team_repository.eats_leads_dining_api, │ on forgejo_team_repos.tf.json line 327, in resource.forgejo_team_repository.eats_leads_dining_api: │ 327: }, │ │ Repository 'ScottyLabs/dining-api' not found in team with ID 27 ╵ ╷ │ Error: Unable to read team repository │ │ with forgejo_team_repository.eats_leads_eats, │ on forgejo_team_repos.tf.json line 336, in resource.forgejo_team_repository.eats_leads_eats: │ 336: }, │ │ Repository 'ScottyLabs/eats' not found in team with ID 27 ╵ ``` </details>
Owner

atlantis state -p git rm forgejo_team_repository.eats_leads_dining_api forgejo_team_repository.eats_leads_eats

atlantis state -p git rm forgejo_team_repository.eats_leads_dining_api forgejo_team_repository.eats_leads_eats

Ran State rm for project: git dir: tofu/git workspace: default

Removed forgejo_team_repository.eats_leads_dining_api
Removed forgejo_team_repository.eats_leads_eats
Successfully removed 2 resource instance(s).

🚮 A plan file was discarded. Re-plan would be required before applying.

  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    
Ran State `rm` for project: `git` dir: `tofu/git` workspace: `default` ```diff Removed forgejo_team_repository.eats_leads_dining_api Removed forgejo_team_repository.eats_leads_eats Successfully removed 2 resource instance(s). ``` :put_litter_in_its_place: A plan file was discarded. Re-plan would be required before applying. * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ```
Owner

atlantis plan -p git

atlantis plan -p git

Ran Plan for project: git dir: tofu/git workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.05s


Note: Objects have changed outside of OpenTofu

OpenTofu detected the following changes made outside of OpenTofu since the
last "tofu apply" which may have affected this plan:

  # restapi_object.cmu-eats_dining_api_push_mirror has been deleted
- resource "restapi_object" "cmu-eats_dining_api_push_mirror" {
-       api_data        = {
          - "public_key"     = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIIKyjdH3wH8J3aSgeZMLB9/uUGzyYltQsk5Cl1Q8MXqF"
            # (9 unchanged elements hidden)
        } -> null
        id              = "remote_mirror_a4MsQ5rO1_Z"
        # (5 unchanged attributes hidden)
    }

  # restapi_object.cmu-eats_eats_push_mirror has been deleted
- resource "restapi_object" "cmu-eats_eats_push_mirror" {
-       api_data        = {
          - "public_key"     = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFsm79Zlyi0JKH4y/BdgCCoLyrC4/18yF3owfsqOt53/"
            # (9 unchanged elements hidden)
        } -> null
        id              = "remote_mirror_sjYwMA6NPgh"
        # (5 unchanged attributes hidden)
    }


Unless you have made equivalent changes to your configuration, or ignored the
relevant attributes using ignore_changes, the following plan may include
actions to undo or respond to these changes.

─────────────────────────────────────────────────────────────────────────────

OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)
-/+ destroy and then create replacement

OpenTofu will perform the following actions:

  # forgejo_repository.cmu-eats_dining_api will be updated in-place
! resource "forgejo_repository" "cmu-eats_dining_api" {
!       clone_url                   = "https://git.cmu.dev/ScottyLabs/dining-api.git" -> (known after apply)
!       created_at                  = "2026-09-18T23:16:27Z" -> (known after apply)
+       description                 = "Dining API for CMU dining locations and availability"
!       empty                       = false -> (known after apply)
+       external_tracker            = (known after apply)
+       external_wiki               = (known after apply)
!       forks_count                 = 0 -> (known after apply)
!       full_name                   = "ScottyLabs/dining-api" -> (known after apply)
!       html_url                    = "https://git.cmu.dev/ScottyLabs/dining-api" -> (known after apply)
        id                          = 97
!       internal_tracker            = {
!           allow_only_contributors_to_track_time = true -> (known after apply)
!           enable_issue_dependencies             = true -> (known after apply)
!           enable_time_tracker                   = true -> (known after apply)
        } -> (known after apply)
+       mirror_interval             = (known after apply)
!       mirror_updated              = "0001-01-01T00:00:00Z" -> (known after apply)
        name                        = "dining-api"
!       open_issues_count           = 0 -> (known after apply)
!       open_pr_counter             = 5 -> (known after apply)
!       permissions                 = {
!           admin = true -> (known after apply)
!           pull  = true -> (known after apply)
!           push  = true -> (known after apply)
        } -> (known after apply)
!       release_counter             = 0 -> (known after apply)
!       size                        = 38816 -> (known after apply)
!       ssh_url                     = "ssh://forgejo@git.cmu.dev/ScottyLabs/dining-api.git" -> (known after apply)
!       stars_count                 = 0 -> (known after apply)
!       updated_at                  = "2026-09-18T23:22:22Z" -> (known after apply)
!       watchers_count              = 3 -> (known after apply)
!       website                     = "https://git.cmu.dev/ScottyLabs/dining-api" -> "https://api.cmueats.com"
        # (30 unchanged attributes hidden)
    }

  # forgejo_repository.cmu-eats_eats will be updated in-place
! resource "forgejo_repository" "cmu-eats_eats" {
!       clone_url                   = "https://git.cmu.dev/ScottyLabs/eats.git" -> (known after apply)
!       created_at                  = "2026-09-18T23:15:39Z" -> (known after apply)
!       description                 = "Vite web app for CMU dining location info" -> "CMU dining location info"
!       empty                       = false -> (known after apply)
+       external_tracker            = (known after apply)
+       external_wiki               = (known after apply)
!       forks_count                 = 0 -> (known after apply)
!       full_name                   = "ScottyLabs/eats" -> (known after apply)
!       html_url                    = "https://git.cmu.dev/ScottyLabs/eats" -> (known after apply)
        id                          = 96
!       internal_tracker            = {
!           allow_only_contributors_to_track_time = true -> (known after apply)
!           enable_issue_dependencies             = true -> (known after apply)
!           enable_time_tracker                   = true -> (known after apply)
        } -> (known after apply)
+       mirror_interval             = (known after apply)
!       mirror_updated              = "0001-01-01T00:00:00Z" -> (known after apply)
        name                        = "eats"
!       open_issues_count           = 48 -> (known after apply)
!       open_pr_counter             = 4 -> (known after apply)
!       permissions                 = {
!           admin = true -> (known after apply)
!           pull  = true -> (known after apply)
!           push  = true -> (known after apply)
        } -> (known after apply)
!       release_counter             = 0 -> (known after apply)
!       size                        = 95496 -> (known after apply)
!       ssh_url                     = "ssh://forgejo@git.cmu.dev/ScottyLabs/eats.git" -> (known after apply)
!       stars_count                 = 0 -> (known after apply)
!       updated_at                  = "2026-09-18T23:35:12Z" -> (known after apply)
!       watchers_count              = 3 -> (known after apply)
        # (31 unchanged attributes hidden)
    }

  # forgejo_repository.cmu-study_study will be created
+ resource "forgejo_repository" "cmu-study_study" {
+       allow_manual_merge          = false
+       allow_merge_commits         = true
+       allow_rebase                = true
+       allow_rebase_explicit       = true
+       allow_squash_merge          = true
+       archive_on_destroy          = false
+       archived                    = false
+       auto_init                   = true
+       autodetect_manual_merge     = false
+       avatar_url                  = (known after apply)
+       clone_url                   = (known after apply)
+       created_at                  = (known after apply)
+       default_branch              = "main"
+       default_merge_style         = "merge"
+       description                 = "A platform for creating and joining study groups"
+       empty                       = (known after apply)
+       external_tracker            = (known after apply)
+       external_wiki               = (known after apply)
+       fork                        = (known after apply)
+       forks_count                 = (known after apply)
+       full_name                   = (known after apply)
+       has_actions                 = true
+       has_issues                  = true
+       has_packages                = true
+       has_projects                = true
+       has_pull_requests           = true
+       has_releases                = true
+       has_wiki                    = true
+       html_url                    = (known after apply)
+       id                          = (known after apply)
+       ignore_whitespace_conflicts = false
+       internal                    = (known after apply)
+       internal_tracker            = (known after apply)
+       labels                      = false
+       lfs                         = false
+       milestones                  = false
+       mirror                      = false
+       mirror_interval             = (known after apply)
+       mirror_updated              = (known after apply)
+       name                        = "study"
+       open_issues_count           = (known after apply)
+       open_pr_counter             = (known after apply)
+       owner                       = "ScottyLabs"
+       parent_id                   = (known after apply)
+       permissions                 = (known after apply)
+       private                     = false
+       release_counter             = (known after apply)
+       size                        = (known after apply)
+       ssh_url                     = (known after apply)
+       stars_count                 = (known after apply)
+       template                    = false
+       trust_model                 = "default"
+       updated_at                  = (known after apply)
+       watchers_count              = (known after apply)
+       website                     = "https://cmustudy.com"
    }

  # forgejo_team.cmu-study will be created
+ resource "forgejo_team" "cmu-study" {
+       can_create_org_repo       = (known after apply)
+       description               = "CMU Study"
+       id                        = (known after apply)
+       includes_all_repositories = (known after apply)
+       name                      = "cmu-study"
+       organization              = (known after apply)
+       organization_id           = 3
+       permission                = "write"
+       units_map                 = {
          + "repo.code"   = "write"
          + "repo.issues" = "write"
          + "repo.pulls"  = "write"
        }
    }

  # forgejo_team.study_leads will be created
+ resource "forgejo_team" "study_leads" {
+       can_create_org_repo       = false
+       description               = "CMU Study leads"
+       id                        = (known after apply)
+       includes_all_repositories = false
+       name                      = "study-leads"
+       organization              = (known after apply)
+       organization_id           = 3
+       permission                = "admin"
+       units_map                 = {
          + "repo.actions"    = "admin"
          + "repo.code"       = "admin"
          + "repo.ext_issues" = "admin"
          + "repo.ext_wiki"   = "admin"
          + "repo.issues"     = "admin"
          + "repo.packages"   = "admin"
          + "repo.projects"   = "admin"
          + "repo.pulls"      = "admin"
          + "repo.releases"   = "admin"
          + "repo.wiki"       = "admin"
        }
    }

  # forgejo_team_member.cmu-study_annadavi will be created
+ resource "forgejo_team_member" "cmu-study_annadavi" {
+       team_id = (known after apply)
+       user    = "annadavi"
    }

  # forgejo_team_member.study_leads_annadavi will be created
+ resource "forgejo_team_member" "study_leads_annadavi" {
+       team_id = (known after apply)
+       user    = "annadavi"
    }

  # forgejo_team_repository.cmu-eats_dining_api will be created
+ resource "forgejo_team_repository" "cmu-eats_dining_api" {
+       owner      = "ScottyLabs"
+       repository = "dining-api"
+       team_id    = 28
    }

  # forgejo_team_repository.cmu-eats_eats will be created
+ resource "forgejo_team_repository" "cmu-eats_eats" {
+       owner      = "ScottyLabs"
+       repository = "eats"
+       team_id    = 28
    }

  # forgejo_team_repository.cmu-study_study will be created
+ resource "forgejo_team_repository" "cmu-study_study" {
+       owner      = "ScottyLabs"
+       repository = "study"
+       team_id    = (known after apply)
    }

  # forgejo_team_repository.eats_leads_dining_api will be created
+ resource "forgejo_team_repository" "eats_leads_dining_api" {
+       owner      = "ScottyLabs"
+       repository = "dining-api"
+       team_id    = 27
    }

  # forgejo_team_repository.eats_leads_eats will be created
+ resource "forgejo_team_repository" "eats_leads_eats" {
+       owner      = "ScottyLabs"
+       repository = "eats"
+       team_id    = 27
    }

  # forgejo_team_repository.study_leads_study will be created
+ resource "forgejo_team_repository" "study_leads_study" {
+       owner      = "ScottyLabs"
+       repository = "study"
+       team_id    = (known after apply)
    }

  # github_actions_repository_permissions.cmu-study_study_actions will be created
+ resource "github_actions_repository_permissions" "cmu-study_study_actions" {
+       enabled              = false
+       id                   = (known after apply)
+       repository           = "study"
+       sha_pinning_required = (known after apply)
    }

  # github_repository.cmu-study_study will be created
+ resource "github_repository" "cmu-study_study" {
+       allow_auto_merge                        = false
+       allow_forking                           = (known after apply)
+       allow_merge_commit                      = true
+       allow_rebase_merge                      = true
+       allow_squash_merge                      = true
+       archived                                = false
+       default_branch                          = (known after apply)
+       delete_branch_on_merge                  = false
+       description                             = "READ-ONLY MIRROR"
+       etag                                    = (known after apply)
+       fork                                    = (known after apply)
+       full_name                               = (known after apply)
+       git_clone_url                           = (known after apply)
+       has_discussions                         = false
+       has_issues                              = false
+       has_projects                            = false
+       has_wiki                                = false
+       homepage_url                            = "https://git.cmu.dev/ScottyLabs/study"
+       html_url                                = (known after apply)
+       http_clone_url                          = (known after apply)
+       id                                      = (known after apply)
+       ignore_vulnerability_alerts_during_read = false
+       merge_commit_message                    = "PR_TITLE"
+       merge_commit_title                      = "MERGE_MESSAGE"
+       name                                    = "study"
+       node_id                                 = (known after apply)
+       primary_language                        = (known after apply)
+       private                                 = (known after apply)
+       repo_id                                 = (known after apply)
+       source_owner                            = (known after apply)
+       source_repo                             = (known after apply)
+       squash_merge_commit_message             = "COMMIT_MESSAGES"
+       squash_merge_commit_title               = "COMMIT_OR_PR_TITLE"
+       ssh_clone_url                           = (known after apply)
+       svn_url                                 = (known after apply)
+       topics                                  = (known after apply)
+       visibility                              = "public"
+       vulnerability_alerts                    = (known after apply)
+       web_commit_signoff_required             = (known after apply)

+       security_and_analysis (known after apply)
    }

  # github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key must be replaced
-/+ resource "github_repository_deploy_key" "cmu-eats_dining_api_mirror_deploy_key" {
!       etag       = "W/\"bf394a0fcc9e1231b4ee544598e6680359d71d177764d9f0c8cd22f5c7ec3ef4\"" -> (known after apply)
!       id         = "dining-api:161772815" -> (known after apply)
!       key        = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIIKyjdH3wH8J3aSgeZMLB9/uUGzyYltQsk5Cl1Q8MXqF" # forces replacement -> (known after apply) # forces replacement
        # (3 unchanged attributes hidden)
    }

  # github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key must be replaced
-/+ resource "github_repository_deploy_key" "cmu-eats_eats_mirror_deploy_key" {
!       etag       = "W/\"5cf445e3b52f1b2e59fe1cb32c5c2e0859e2598566a7aa75c3ebdbf4b01f6164\"" -> (known after apply)
!       id         = "eats:161775995" -> (known after apply)
!       key        = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFsm79Zlyi0JKH4y/BdgCCoLyrC4/18yF3owfsqOt53/" # forces replacement -> (known after apply) # forces replacement
        # (3 unchanged attributes hidden)
    }

  # github_repository_deploy_key.cmu-study_study_mirror_deploy_key will be created
+ resource "github_repository_deploy_key" "cmu-study_study_mirror_deploy_key" {
+       etag       = (known after apply)
+       id         = (known after apply)
+       key        = (known after apply)
+       read_only  = false
+       repository = "study"
+       title      = "Forgejo mirror"
    }

  # github_repository_ruleset.cmu-eats_dining_api_ruleset will be updated in-place
! resource "github_repository_ruleset" "cmu-eats_dining_api_ruleset" {
        id          = "21879626"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 161772815 -> (known after apply)
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place
! resource "github_repository_ruleset" "cmu-eats_eats_ruleset" {
        id          = "21330218"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 161775995 -> (known after apply)
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.cmu-study_study_ruleset will be created
+ resource "github_repository_ruleset" "cmu-study_study_ruleset" {
+       enforcement = "active"
+       etag        = (known after apply)
+       id          = (known after apply)
+       name        = "Default"
+       node_id     = (known after apply)
+       repository  = "study"
+       ruleset_id  = (known after apply)
+       target      = "branch"

+       bypass_actors {
+           actor_id    = (known after apply)
+           actor_type  = "DeployKey"
+           bypass_mode = "always"
        }

+       conditions {
+           ref_name {
+               exclude = []
+               include = [
+                   "~ALL",
                ]
            }
        }

+       rules {
+           creation                      = true
+           deletion                      = true
+           non_fast_forward              = true
+           update                        = true
+           update_allows_fetch_and_merge = false

+           pull_request {
+               allowed_merge_methods             = (known after apply)
+               dismiss_stale_reviews_on_push     = false
+               require_code_owner_review         = false
+               require_last_push_approval        = false
+               required_approving_review_count   = 1
+               required_review_thread_resolution = false
            }
        }
    }

  # github_team.cmu-study will be created
+ resource "github_team" "cmu-study" {
+       create_default_maintainer = false
+       etag                      = (known after apply)
+       id                        = (known after apply)
+       members_count             = (known after apply)
+       name                      = "CMU Study"
+       node_id                   = (known after apply)
+       notification_setting      = "notifications_enabled"
+       parent_team_read_id       = (known after apply)
+       parent_team_read_slug     = (known after apply)
+       privacy                   = "closed"
+       slug                      = (known after apply)
    }

  # github_team_membership.cmu-study_annadavi will be created
+ resource "github_team_membership" "cmu-study_annadavi" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "maintainer"
+       team_id  = (known after apply)
+       username = "annadavid1"
    }

  # restapi_object.cmu-eats_dining_api_push_mirror will be created
+ resource "restapi_object" "cmu-eats_dining_api_push_mirror" {
+       api_data        = (known after apply)
+       api_response    = (known after apply)
+       create_response = (known after apply)
+       data            = jsonencode(
            {
+               interval       = "8h0m0s"
+               remote_address = "git@github.com:ScottyLabs/dining-api.git"
+               sync_on_commit = true
+               use_ssh        = true
            }
        )
+       id              = (known after apply)
+       id_attribute    = "remote_name"
+       path            = "/api/v1/repos/ScottyLabs/dining-api/push_mirrors"
    }

  # restapi_object.cmu-eats_eats_kennel_webhook will be created
+ resource "restapi_object" "cmu-eats_eats_kennel_webhook" {
+       api_data        = (known after apply)
+       api_response    = (known after apply)
+       create_response = (known after apply)
+       data            = (sensitive value)
+       id              = (known after apply)
+       id_attribute    = "id"
+       path            = "/api/v1/repos/ScottyLabs/eats/hooks"
+       update_method   = "PATCH"
    }

  # restapi_object.cmu-eats_eats_push_mirror will be created
+ resource "restapi_object" "cmu-eats_eats_push_mirror" {
+       api_data        = (known after apply)
+       api_response    = (known after apply)
+       create_response = (known after apply)
+       data            = jsonencode(
            {
+               interval       = "8h0m0s"
+               remote_address = "git@github.com:ScottyLabs/eats.git"
+               sync_on_commit = true
+               use_ssh        = true
            }
        )
+       id              = (known after apply)
+       id_attribute    = "remote_name"
+       path            = "/api/v1/repos/ScottyLabs/eats/push_mirrors"
    }

  # restapi_object.cmu-study_study_kennel_webhook will be created
+ resource "restapi_object" "cmu-study_study_kennel_webhook" {
+       api_data        = (known after apply)
+       api_response    = (known after apply)
+       create_response = (known after apply)
+       data            = (sensitive value)
+       id              = (known after apply)
+       id_attribute    = "id"
+       path            = "/api/v1/repos/ScottyLabs/study/hooks"
+       update_method   = "PATCH"
    }

  # restapi_object.cmu-study_study_push_mirror will be created
+ resource "restapi_object" "cmu-study_study_push_mirror" {
+       api_data        = (known after apply)
+       api_response    = (known after apply)
+       create_response = (known after apply)
+       data            = jsonencode(
            {
+               interval       = "8h0m0s"
+               remote_address = "git@github.com:ScottyLabs/study.git"
+               sync_on_commit = true
+               use_ssh        = true
            }
        )
+       id              = (known after apply)
+       id_attribute    = "remote_name"
+       path            = "/api/v1/repos/ScottyLabs/study/push_mirrors"
    }

Plan: 24 to add, 4 to change, 2 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 24 to add, 4 to change, 2 to destroy.


  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.05s Note: Objects have changed outside of OpenTofu OpenTofu detected the following changes made outside of OpenTofu since the last "tofu apply" which may have affected this plan: # restapi_object.cmu-eats_dining_api_push_mirror has been deleted - resource "restapi_object" "cmu-eats_dining_api_push_mirror" { - api_data = { - "public_key" = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIIKyjdH3wH8J3aSgeZMLB9/uUGzyYltQsk5Cl1Q8MXqF" # (9 unchanged elements hidden) } -> null id = "remote_mirror_a4MsQ5rO1_Z" # (5 unchanged attributes hidden) } # restapi_object.cmu-eats_eats_push_mirror has been deleted - resource "restapi_object" "cmu-eats_eats_push_mirror" { - api_data = { - "public_key" = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFsm79Zlyi0JKH4y/BdgCCoLyrC4/18yF3owfsqOt53/" # (9 unchanged elements hidden) } -> null id = "remote_mirror_sjYwMA6NPgh" # (5 unchanged attributes hidden) } Unless you have made equivalent changes to your configuration, or ignored the relevant attributes using ignore_changes, the following plan may include actions to undo or respond to these changes. ───────────────────────────────────────────────────────────────────────────── OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) -/+ destroy and then create replacement OpenTofu will perform the following actions: # forgejo_repository.cmu-eats_dining_api will be updated in-place ! resource "forgejo_repository" "cmu-eats_dining_api" { ! clone_url = "https://git.cmu.dev/ScottyLabs/dining-api.git" -> (known after apply) ! created_at = "2026-09-18T23:16:27Z" -> (known after apply) + description = "Dining API for CMU dining locations and availability" ! empty = false -> (known after apply) + external_tracker = (known after apply) + external_wiki = (known after apply) ! forks_count = 0 -> (known after apply) ! full_name = "ScottyLabs/dining-api" -> (known after apply) ! html_url = "https://git.cmu.dev/ScottyLabs/dining-api" -> (known after apply) id = 97 ! internal_tracker = { ! allow_only_contributors_to_track_time = true -> (known after apply) ! enable_issue_dependencies = true -> (known after apply) ! enable_time_tracker = true -> (known after apply) } -> (known after apply) + mirror_interval = (known after apply) ! mirror_updated = "0001-01-01T00:00:00Z" -> (known after apply) name = "dining-api" ! open_issues_count = 0 -> (known after apply) ! open_pr_counter = 5 -> (known after apply) ! permissions = { ! admin = true -> (known after apply) ! pull = true -> (known after apply) ! push = true -> (known after apply) } -> (known after apply) ! release_counter = 0 -> (known after apply) ! size = 38816 -> (known after apply) ! ssh_url = "ssh://forgejo@git.cmu.dev/ScottyLabs/dining-api.git" -> (known after apply) ! stars_count = 0 -> (known after apply) ! updated_at = "2026-09-18T23:22:22Z" -> (known after apply) ! watchers_count = 3 -> (known after apply) ! website = "https://git.cmu.dev/ScottyLabs/dining-api" -> "https://api.cmueats.com" # (30 unchanged attributes hidden) } # forgejo_repository.cmu-eats_eats will be updated in-place ! resource "forgejo_repository" "cmu-eats_eats" { ! clone_url = "https://git.cmu.dev/ScottyLabs/eats.git" -> (known after apply) ! created_at = "2026-09-18T23:15:39Z" -> (known after apply) ! description = "Vite web app for CMU dining location info" -> "CMU dining location info" ! empty = false -> (known after apply) + external_tracker = (known after apply) + external_wiki = (known after apply) ! forks_count = 0 -> (known after apply) ! full_name = "ScottyLabs/eats" -> (known after apply) ! html_url = "https://git.cmu.dev/ScottyLabs/eats" -> (known after apply) id = 96 ! internal_tracker = { ! allow_only_contributors_to_track_time = true -> (known after apply) ! enable_issue_dependencies = true -> (known after apply) ! enable_time_tracker = true -> (known after apply) } -> (known after apply) + mirror_interval = (known after apply) ! mirror_updated = "0001-01-01T00:00:00Z" -> (known after apply) name = "eats" ! open_issues_count = 48 -> (known after apply) ! open_pr_counter = 4 -> (known after apply) ! permissions = { ! admin = true -> (known after apply) ! pull = true -> (known after apply) ! push = true -> (known after apply) } -> (known after apply) ! release_counter = 0 -> (known after apply) ! size = 95496 -> (known after apply) ! ssh_url = "ssh://forgejo@git.cmu.dev/ScottyLabs/eats.git" -> (known after apply) ! stars_count = 0 -> (known after apply) ! updated_at = "2026-09-18T23:35:12Z" -> (known after apply) ! watchers_count = 3 -> (known after apply) # (31 unchanged attributes hidden) } # forgejo_repository.cmu-study_study will be created + resource "forgejo_repository" "cmu-study_study" { + allow_manual_merge = false + allow_merge_commits = true + allow_rebase = true + allow_rebase_explicit = true + allow_squash_merge = true + archive_on_destroy = false + archived = false + auto_init = true + autodetect_manual_merge = false + avatar_url = (known after apply) + clone_url = (known after apply) + created_at = (known after apply) + default_branch = "main" + default_merge_style = "merge" + description = "A platform for creating and joining study groups" + empty = (known after apply) + external_tracker = (known after apply) + external_wiki = (known after apply) + fork = (known after apply) + forks_count = (known after apply) + full_name = (known after apply) + has_actions = true + has_issues = true + has_packages = true + has_projects = true + has_pull_requests = true + has_releases = true + has_wiki = true + html_url = (known after apply) + id = (known after apply) + ignore_whitespace_conflicts = false + internal = (known after apply) + internal_tracker = (known after apply) + labels = false + lfs = false + milestones = false + mirror = false + mirror_interval = (known after apply) + mirror_updated = (known after apply) + name = "study" + open_issues_count = (known after apply) + open_pr_counter = (known after apply) + owner = "ScottyLabs" + parent_id = (known after apply) + permissions = (known after apply) + private = false + release_counter = (known after apply) + size = (known after apply) + ssh_url = (known after apply) + stars_count = (known after apply) + template = false + trust_model = "default" + updated_at = (known after apply) + watchers_count = (known after apply) + website = "https://cmustudy.com" } # forgejo_team.cmu-study will be created + resource "forgejo_team" "cmu-study" { + can_create_org_repo = (known after apply) + description = "CMU Study" + id = (known after apply) + includes_all_repositories = (known after apply) + name = "cmu-study" + organization = (known after apply) + organization_id = 3 + permission = "write" + units_map = { + "repo.code" = "write" + "repo.issues" = "write" + "repo.pulls" = "write" } } # forgejo_team.study_leads will be created + resource "forgejo_team" "study_leads" { + can_create_org_repo = false + description = "CMU Study leads" + id = (known after apply) + includes_all_repositories = false + name = "study-leads" + organization = (known after apply) + organization_id = 3 + permission = "admin" + units_map = { + "repo.actions" = "admin" + "repo.code" = "admin" + "repo.ext_issues" = "admin" + "repo.ext_wiki" = "admin" + "repo.issues" = "admin" + "repo.packages" = "admin" + "repo.projects" = "admin" + "repo.pulls" = "admin" + "repo.releases" = "admin" + "repo.wiki" = "admin" } } # forgejo_team_member.cmu-study_annadavi will be created + resource "forgejo_team_member" "cmu-study_annadavi" { + team_id = (known after apply) + user = "annadavi" } # forgejo_team_member.study_leads_annadavi will be created + resource "forgejo_team_member" "study_leads_annadavi" { + team_id = (known after apply) + user = "annadavi" } # forgejo_team_repository.cmu-eats_dining_api will be created + resource "forgejo_team_repository" "cmu-eats_dining_api" { + owner = "ScottyLabs" + repository = "dining-api" + team_id = 28 } # forgejo_team_repository.cmu-eats_eats will be created + resource "forgejo_team_repository" "cmu-eats_eats" { + owner = "ScottyLabs" + repository = "eats" + team_id = 28 } # forgejo_team_repository.cmu-study_study will be created + resource "forgejo_team_repository" "cmu-study_study" { + owner = "ScottyLabs" + repository = "study" + team_id = (known after apply) } # forgejo_team_repository.eats_leads_dining_api will be created + resource "forgejo_team_repository" "eats_leads_dining_api" { + owner = "ScottyLabs" + repository = "dining-api" + team_id = 27 } # forgejo_team_repository.eats_leads_eats will be created + resource "forgejo_team_repository" "eats_leads_eats" { + owner = "ScottyLabs" + repository = "eats" + team_id = 27 } # forgejo_team_repository.study_leads_study will be created + resource "forgejo_team_repository" "study_leads_study" { + owner = "ScottyLabs" + repository = "study" + team_id = (known after apply) } # github_actions_repository_permissions.cmu-study_study_actions will be created + resource "github_actions_repository_permissions" "cmu-study_study_actions" { + enabled = false + id = (known after apply) + repository = "study" + sha_pinning_required = (known after apply) } # github_repository.cmu-study_study will be created + resource "github_repository" "cmu-study_study" { + allow_auto_merge = false + allow_forking = (known after apply) + allow_merge_commit = true + allow_rebase_merge = true + allow_squash_merge = true + archived = false + default_branch = (known after apply) + delete_branch_on_merge = false + description = "READ-ONLY MIRROR" + etag = (known after apply) + fork = (known after apply) + full_name = (known after apply) + git_clone_url = (known after apply) + has_discussions = false + has_issues = false + has_projects = false + has_wiki = false + homepage_url = "https://git.cmu.dev/ScottyLabs/study" + html_url = (known after apply) + http_clone_url = (known after apply) + id = (known after apply) + ignore_vulnerability_alerts_during_read = false + merge_commit_message = "PR_TITLE" + merge_commit_title = "MERGE_MESSAGE" + name = "study" + node_id = (known after apply) + primary_language = (known after apply) + private = (known after apply) + repo_id = (known after apply) + source_owner = (known after apply) + source_repo = (known after apply) + squash_merge_commit_message = "COMMIT_MESSAGES" + squash_merge_commit_title = "COMMIT_OR_PR_TITLE" + ssh_clone_url = (known after apply) + svn_url = (known after apply) + topics = (known after apply) + visibility = "public" + vulnerability_alerts = (known after apply) + web_commit_signoff_required = (known after apply) + security_and_analysis (known after apply) } # github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key must be replaced -/+ resource "github_repository_deploy_key" "cmu-eats_dining_api_mirror_deploy_key" { ! etag = "W/\"bf394a0fcc9e1231b4ee544598e6680359d71d177764d9f0c8cd22f5c7ec3ef4\"" -> (known after apply) ! id = "dining-api:161772815" -> (known after apply) ! key = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIIKyjdH3wH8J3aSgeZMLB9/uUGzyYltQsk5Cl1Q8MXqF" # forces replacement -> (known after apply) # forces replacement # (3 unchanged attributes hidden) } # github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key must be replaced -/+ resource "github_repository_deploy_key" "cmu-eats_eats_mirror_deploy_key" { ! etag = "W/\"5cf445e3b52f1b2e59fe1cb32c5c2e0859e2598566a7aa75c3ebdbf4b01f6164\"" -> (known after apply) ! id = "eats:161775995" -> (known after apply) ! key = "ssh-ed25519 AAAAC3NzaC1lZDI1NTE5AAAAIFsm79Zlyi0JKH4y/BdgCCoLyrC4/18yF3owfsqOt53/" # forces replacement -> (known after apply) # forces replacement # (3 unchanged attributes hidden) } # github_repository_deploy_key.cmu-study_study_mirror_deploy_key will be created + resource "github_repository_deploy_key" "cmu-study_study_mirror_deploy_key" { + etag = (known after apply) + id = (known after apply) + key = (known after apply) + read_only = false + repository = "study" + title = "Forgejo mirror" } # github_repository_ruleset.cmu-eats_dining_api_ruleset will be updated in-place ! resource "github_repository_ruleset" "cmu-eats_dining_api_ruleset" { id = "21879626" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 161772815 -> (known after apply) # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.cmu-eats_eats_ruleset will be updated in-place ! resource "github_repository_ruleset" "cmu-eats_eats_ruleset" { id = "21330218" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 161775995 -> (known after apply) # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.cmu-study_study_ruleset will be created + resource "github_repository_ruleset" "cmu-study_study_ruleset" { + enforcement = "active" + etag = (known after apply) + id = (known after apply) + name = "Default" + node_id = (known after apply) + repository = "study" + ruleset_id = (known after apply) + target = "branch" + bypass_actors { + actor_id = (known after apply) + actor_type = "DeployKey" + bypass_mode = "always" } + conditions { + ref_name { + exclude = [] + include = [ + "~ALL", ] } } + rules { + creation = true + deletion = true + non_fast_forward = true + update = true + update_allows_fetch_and_merge = false + pull_request { + allowed_merge_methods = (known after apply) + dismiss_stale_reviews_on_push = false + require_code_owner_review = false + require_last_push_approval = false + required_approving_review_count = 1 + required_review_thread_resolution = false } } } # github_team.cmu-study will be created + resource "github_team" "cmu-study" { + create_default_maintainer = false + etag = (known after apply) + id = (known after apply) + members_count = (known after apply) + name = "CMU Study" + node_id = (known after apply) + notification_setting = "notifications_enabled" + parent_team_read_id = (known after apply) + parent_team_read_slug = (known after apply) + privacy = "closed" + slug = (known after apply) } # github_team_membership.cmu-study_annadavi will be created + resource "github_team_membership" "cmu-study_annadavi" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = (known after apply) + username = "annadavid1" } # restapi_object.cmu-eats_dining_api_push_mirror will be created + resource "restapi_object" "cmu-eats_dining_api_push_mirror" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = jsonencode( { + interval = "8h0m0s" + remote_address = "git@github.com:ScottyLabs/dining-api.git" + sync_on_commit = true + use_ssh = true } ) + id = (known after apply) + id_attribute = "remote_name" + path = "/api/v1/repos/ScottyLabs/dining-api/push_mirrors" } # restapi_object.cmu-eats_eats_kennel_webhook will be created + resource "restapi_object" "cmu-eats_eats_kennel_webhook" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = (sensitive value) + id = (known after apply) + id_attribute = "id" + path = "/api/v1/repos/ScottyLabs/eats/hooks" + update_method = "PATCH" } # restapi_object.cmu-eats_eats_push_mirror will be created + resource "restapi_object" "cmu-eats_eats_push_mirror" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = jsonencode( { + interval = "8h0m0s" + remote_address = "git@github.com:ScottyLabs/eats.git" + sync_on_commit = true + use_ssh = true } ) + id = (known after apply) + id_attribute = "remote_name" + path = "/api/v1/repos/ScottyLabs/eats/push_mirrors" } # restapi_object.cmu-study_study_kennel_webhook will be created + resource "restapi_object" "cmu-study_study_kennel_webhook" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = (sensitive value) + id = (known after apply) + id_attribute = "id" + path = "/api/v1/repos/ScottyLabs/study/hooks" + update_method = "PATCH" } # restapi_object.cmu-study_study_push_mirror will be created + resource "restapi_object" "cmu-study_study_push_mirror" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = jsonencode( { + interval = "8h0m0s" + remote_address = "git@github.com:ScottyLabs/study.git" + sync_on_commit = true + use_ssh = true } ) + id = (known after apply) + id_attribute = "remote_name" + path = "/api/v1/repos/ScottyLabs/study/push_mirrors" } Plan: 24 to add, 4 to change, 2 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 24 to add, 4 to change, 2 to destroy. --- * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis apply

atlantis apply

Ran Apply for 10 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: openbao dir: tofu/openbao workspace: default
  4. project: posthog dir: tofu/posthog workspace: default
  5. project: google_groups dir: tofu/google_groups workspace: default
  6. project: sentry dir: tofu/sentry workspace: default
  7. project: keycloak dir: tofu/keycloak workspace: default
  8. project: discord dir: tofu/discord workspace: default
  9. project: matrix_bridges dir: tofu/matrix_bridges workspace: default
  10. project: git dir: tofu/git workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.38s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

2. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.26s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_cmu-study_ch0_annadavi: Creating...
null_resource.slack_leads_annadavi: Creating...
null_resource.slack_hub_annadavi: Creating...
null_resource.slack_cmu-study_ch0_annadavi: Provisioning with 'local-exec'...
null_resource.slack_cmu-study_ch0_annadavi (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C09KN6M3P53 --user U07M24AKRGV"]
null_resource.slack_leads_annadavi: Provisioning with 'local-exec'...
null_resource.slack_leads_annadavi (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0AN1M0HDV0 --user U07M24AKRGV"]
null_resource.slack_hub_annadavi: Provisioning with 'local-exec'...
null_resource.slack_hub_annadavi (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U07M24AKRGV"]
null_resource.slack_hub_annadavi: Creation complete after 1s [id=968070886944939043]
null_resource.slack_cmu-study_ch0_annadavi: Creation complete after 1s [id=180323051751415711]
null_resource.slack_leads_annadavi: Creation complete after 1s [id=4222181416230461163]

Apply complete! Resources: 3 added, 0 changed, 0 destroyed.

3. project: openbao dir: tofu/openbao workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.29s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
vault_identity_group.study_admins: Creating...
vault_identity_group.cmugpt-surface_admins: Modifying... [id=3ae18ef3-f370-f163-767a-accfb092ea0d]
vault_identity_group.discord-verify_admins: Modifying... [id=c3e29b3b-0ce5-61a3-3a69-e950f297fa78]
vault_identity_group.changelog_admins: Modifying... [id=d73c0ef4-a7d2-d171-ebd1-1c15cee174c8]
vault_identity_group.eats_admins: Modifying... [id=05959815-c896-9703-32eb-7ba84604303c]
vault_identity_group.cal_admins: Modifying... [id=da96ea82-8537-f0a1-0d42-a7ab1cc3c006]
vault_identity_group.components_admins: Modifying... [id=d2971366-64a4-984e-1e4e-dad3cd07c841]
vault_identity_group.housing_admins: Modifying... [id=7c688f49-59f1-1593-1e19-16f70b86eae2]
vault_identity_group.insta_admins: Modifying... [id=be24e607-e8d5-9fa3-a11a-9a8ca05ede04]
vault_identity_group.scottylabs-org_admins: Modifying... [id=0dbd914d-b96c-de7b-1ec5-a0ef330b20e0]
vault_identity_group.cal_admins: Modifications complete after 0s [id=da96ea82-8537-f0a1-0d42-a7ab1cc3c006]
vault_identity_group.cmugpt-surface_admins: Modifications complete after 0s [id=3ae18ef3-f370-f163-767a-accfb092ea0d]
vault_identity_group.scottylabs-org_admins: Modifications complete after 0s [id=0dbd914d-b96c-de7b-1ec5-a0ef330b20e0]
vault_identity_group.changelog_admins: Modifications complete after 0s [id=d73c0ef4-a7d2-d171-ebd1-1c15cee174c8]
vault_identity_group.study_admins: Creation complete after 0s [id=14e88827-391c-ab7b-acfa-b824a25f0c4a]
vault_identity_group.discord-verify_admins: Modifications complete after 0s [id=c3e29b3b-0ce5-61a3-3a69-e950f297fa78]
vault_identity_group.housing_admins: Modifications complete after 0s [id=7c688f49-59f1-1593-1e19-16f70b86eae2]
vault_identity_group.components_admins: Modifications complete after 0s [id=d2971366-64a4-984e-1e4e-dad3cd07c841]
vault_identity_group.occ-main_admins: Modifying... [id=4d40db70-1bac-505e-ec8d-5a9d5cf15ac7]
vault_identity_group.insta_admins: Modifications complete after 0s [id=be24e607-e8d5-9fa3-a11a-9a8ca05ede04]
vault_identity_group.eats_admins: Modifications complete after 0s [id=05959815-c896-9703-32eb-7ba84604303c]
vault_policy.study_prod: Creating...
vault_identity_group.study_members: Creating...
vault_identity_group.cmugpt-agent_admins: Modifying... [id=cb081cd2-a3cd-81e1-8c20-0914cac5c213]
vault_identity_group.maps_admins: Modifying... [id=6b2bd85e-e311-ae83-8a07-30100f0ac12f]
vault_policy.study_dev: Creating...
vault_identity_group.lost-and-found-v2_admins: Modifying... [id=db8b5072-bb86-c9d0-e93f-951c2901bd15]
vault_identity_group.cmugpt-mcp-server_admins: Modifying... [id=a4717a6f-015f-9fae-c6a2-76657b8a7464]
vault_identity_group.quest_admins: Modifying... [id=c03f2eab-1701-f315-fc17-39814ddcb34b]
vault_identity_group.groupme-mirror_admins: Modifying... [id=e264aabf-8766-67fd-f0bf-860a240ee2c2]
vault_identity_group.occ-main_admins: Modifications complete after 1s [id=4d40db70-1bac-505e-ec8d-5a9d5cf15ac7]
vault_identity_group.documentation_admins: Modifying... [id=35c5df83-0e17-31fc-7b82-3365d5526dd5]
vault_identity_group.maps_admins: Modifications complete after 1s [id=6b2bd85e-e311-ae83-8a07-30100f0ac12f]
vault_identity_group.courses_admins: Modifying... [id=d090ae17-1992-86a5-1f8b-092c937092e8]
vault_identity_group.lost-and-found-v2_admins: Modifications complete after 1s [id=db8b5072-bb86-c9d0-e93f-951c2901bd15]
vault_identity_group.study_members: Creation complete after 1s [id=d9c27753-d898-8113-e4cd-1b3fb4b8f73b]
vault_identity_group.cmugpt-agent_admins: Modifications complete after 1s [id=cb081cd2-a3cd-81e1-8c20-0914cac5c213]
vault_identity_group.link-shortener_admins: Modifying... [id=d3fa8791-4bd4-db20-68b6-79db5ac36a25]
vault_identity_group.bus-sign_admins: Modifying... [id=1b72d8b5-8511-fe92-c98c-2ddb8cbf17a1]
vault_identity_group.cmugpt-mcp-server_admins: Modifications complete after 1s [id=a4717a6f-015f-9fae-c6a2-76657b8a7464]
vault_identity_group.quest_admins: Modifications complete after 0s [id=c03f2eab-1701-f315-fc17-39814ddcb34b]
vault_identity_group.tartan-vote_admins: Modifying... [id=ff77ad01-20b1-c320-37bd-6185b3f98d80]
vault_identity_group.groupme-mirror_admins: Modifications complete after 0s [id=e264aabf-8766-67fd-f0bf-860a240ee2c2]
vault_identity_group.terrier_admins: Modifying... [id=956692a1-84f6-c95c-6d40-2382ead13476]
vault_identity_group.dalmatian_admins: Modifying... [id=b1d88b18-8b44-c2ea-9220-fc53ee7e3c7e]
vault_identity_group.cmugpt-sms-surface_admins: Modifying... [id=780f9498-dee7-8cac-9bbc-e1feab5ea2fb]
vault_policy.study_dev: Creation complete after 1s [id=study-dev]
vault_policy.study_prod: Creation complete after 1s [id=study-prod]
vault_identity_group.documentation_admins: Modifications complete after 0s [id=35c5df83-0e17-31fc-7b82-3365d5526dd5]
vault_identity_group_alias.study_admins: Creating...
vault_identity_group_alias.study_members: Creating...
vault_identity_group.courses_admins: Modifications complete after 0s [id=d090ae17-1992-86a5-1f8b-092c937092e8]
vault_identity_group.link-shortener_admins: Modifications complete after 0s [id=d3fa8791-4bd4-db20-68b6-79db5ac36a25]
vault_identity_group.bus-sign_admins: Modifications complete after 0s [id=1b72d8b5-8511-fe92-c98c-2ddb8cbf17a1]
vault_identity_group.tartan-vote_admins: Modifications complete after 0s [id=ff77ad01-20b1-c320-37bd-6185b3f98d80]
vault_identity_group.cmugpt-sms-surface_admins: Modifications complete after 0s [id=780f9498-dee7-8cac-9bbc-e1feab5ea2fb]
vault_identity_group.dalmatian_admins: Modifications complete after 0s [id=b1d88b18-8b44-c2ea-9220-fc53ee7e3c7e]
vault_identity_group.terrier_admins: Modifications complete after 0s [id=956692a1-84f6-c95c-6d40-2382ead13476]
vault_identity_group_alias.study_admins: Creation complete after 0s [id=f515d97d-daff-e0c3-f762-00a95eb26ea3]
vault_identity_group_alias.study_members: Creation complete after 0s [id=4426b35a-0210-225b-1afb-b79089f12a41]

Apply complete! Resources: 6 added, 24 changed, 0 destroyed.

4. project: posthog dir: tofu/posthog workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.47s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
posthog_project.cmu-study_study: Creating...
vault_kv_secret_v2.cmu-study_study_posthog_host: Creating...
vault_kv_secret_v2.cmu-study_study_posthog_host: Creation complete after 0s [id=secret/data/secretspec/study/prod/POSTHOG_HOST]
posthog_project.cmu-study_study: Creation complete after 1s [name=study]
vault_kv_secret_v2.cmu-study_study_posthog_key: Creating...
vault_kv_secret_v2.cmu-study_study_posthog_key: Creation complete after 0s [id=secret/data/secretspec/study/prod/POSTHOG_KEY]

Apply complete! Resources: 3 added, 0 changed, 0 destroyed.

5. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.53s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_annadavi: Creating...
google_cloud_identity_group_membership.ops_annadavi: Creating...
google_cloud_identity_group_membership.ops_annadavi: Creation complete after 1s [id=groups/03fwokq036zsp1f/memberships/113975979101831473222]
google_cloud_identity_group_membership.tech_annadavi: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/113975979101831473222]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

6. project: sentry dir: tofu/sentry workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.37s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
sentry_team.cmu-study: Creating...
sentry_team.cmu-study: Creation complete after 0s [id=cmu-study]
sentry_project.cmu-study_study: Creating...
sentry_project.cmu-study_study: Creation complete after 2s [id=study]
sentry_key.cmu-study_study: Creating...
sentry_key.cmu-study_study: Creation complete after 0s [id=e1d434c547505c7213a3b2618030187c]
vault_kv_secret_v2.cmu-study_study_sentry_dsn: Creating...
vault_kv_secret_v2.cmu-study_study_sentry_dsn: Creation complete after 1s [id=secret/data/secretspec/study/prod/SENTRY_DSN]

Apply complete! Resources: 4 added, 0 changed, 0 destroyed.

7. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.39s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group.project_cmu-study_study: Creating...
keycloak_openid_client.study_dev: Creating...
keycloak_openid_client.study_staging: Creating...
keycloak_openid_client.study: Creating...
vault_kv_secret_v2.study_preview_keycloak_realm: Creating...
vault_kv_secret_v2.study_staging_keycloak_url: Creating...
vault_kv_secret_v2.study_staging_project_admin_group: Creating...
vault_kv_secret_v2.study_dev_project_admin_group: Creating...
vault_kv_secret_v2.study_dev_oauth_relay_url: Creating...
vault_kv_secret_v2.study_staging_project_group: Creating...
keycloak_group.project_cmu-study_study: Creation complete after 0s [id=f1b424fe-27f7-40ef-831f-9885f66bac3a]
vault_kv_secret_v2.study_prod_keycloak_realm: Creating...
keycloak_openid_client.study_dev: Creation complete after 0s [id=1983432b-c99e-4470-9a59-1f8fdb9f8bef]
vault_kv_secret_v2.study_prod_oauth_relay_url: Creating...
keycloak_openid_client.study_staging: Creation complete after 0s [id=18b84a0e-d9fa-4e99-9456-db6fba3358ff]
keycloak_openid_client.study: Creation complete after 0s [id=7622bc88-d377-4e9b-8995-86ec8e6f4638]
vault_kv_secret_v2.study_prod_keycloak_url: Creating...
vault_kv_secret_v2.study_preview_project_admin_group: Creating...
vault_kv_secret_v2.study_dev_project_admin_group: Creation complete after 0s [id=secret/data/secretspec/study/dev/PROJECT_ADMIN_GROUP]
vault_kv_secret_v2.study_staging_project_group: Creation complete after 0s [id=secret/data/secretspec/study/staging/PROJECT_GROUP]
vault_kv_secret_v2.study_staging_project_admin_group: Creation complete after 0s [id=secret/data/secretspec/study/staging/PROJECT_ADMIN_GROUP]
vault_kv_secret_v2.study_dev_oauth_relay_url: Creation complete after 0s [id=secret/data/secretspec/study/dev/OAUTH_RELAY_URL]
vault_kv_secret_v2.study_staging_keycloak_url: Creation complete after 0s [id=secret/data/secretspec/study/staging/KEYCLOAK_URL]
vault_kv_secret_v2.study_preview_keycloak_realm: Creation complete after 0s [id=secret/data/secretspec/study/preview/KEYCLOAK_REALM]
vault_kv_secret_v2.study_prod_keycloak_realm: Creation complete after 0s [id=secret/data/secretspec/study/prod/KEYCLOAK_REALM]
vault_kv_secret_v2.study_staging_keycloak_realm: Creating...
vault_kv_secret_v2.study_dev_keycloak_url: Creating...
vault_kv_secret_v2.study_dev_keycloak_realm: Creating...
vault_kv_secret_v2.study_staging_oauth_relay_url: Creating...
vault_kv_secret_v2.study_prod_project_group: Creating...
vault_kv_secret_v2.study_prod_project_admin_group: Creating...
vault_kv_secret_v2.study_preview_oauth_relay_url: Creating...
vault_kv_secret_v2.study_prod_keycloak_url: Creation complete after 1s [id=secret/data/secretspec/study/prod/KEYCLOAK_URL]
vault_kv_secret_v2.study_preview_keycloak_url: Creating...
vault_kv_secret_v2.study_prod_oauth_relay_url: Creation complete after 1s [id=secret/data/secretspec/study/prod/OAUTH_RELAY_URL]
vault_kv_secret_v2.study_preview_project_group: Creating...
vault_kv_secret_v2.study_preview_project_admin_group: Creation complete after 1s [id=secret/data/secretspec/study/preview/PROJECT_ADMIN_GROUP]
vault_kv_secret_v2.study_dev_project_group: Creating...
vault_kv_secret_v2.study_staging_oauth_relay_url: Creation complete after 1s [id=secret/data/secretspec/study/staging/OAUTH_RELAY_URL]
vault_kv_secret_v2.study_staging_keycloak_realm: Creation complete after 1s [id=secret/data/secretspec/study/staging/KEYCLOAK_REALM]
vault_kv_secret_v2.study_preview_oauth_relay_url: Creation complete after 1s [id=secret/data/secretspec/study/preview/OAUTH_RELAY_URL]
vault_kv_secret_v2.study_dev_keycloak_url: Creation complete after 1s [id=secret/data/secretspec/study/dev/KEYCLOAK_URL]
vault_kv_secret_v2.study_dev_keycloak_realm: Creation complete after 1s [id=secret/data/secretspec/study/dev/KEYCLOAK_REALM]
keycloak_openid_group_membership_protocol_mapper.study_dev: Creating...
keycloak_group.project_cmu-study_study_admins: Creating...
vault_kv_secret_v2.study_prod_project_admin_group: Creation complete after 1s [id=secret/data/secretspec/study/prod/PROJECT_ADMIN_GROUP]
keycloak_group_memberships.project_cmu-study_study: Creating...
vault_kv_secret_v2.study_prod_project_group: Creation complete after 1s [id=secret/data/secretspec/study/prod/PROJECT_GROUP]
vault_kv_secret_v2.study_dev_oidc_client_secret: Creating...
vault_kv_secret_v2.study_dev_oidc_client_id: Creating...
vault_kv_secret_v2.study_prod_oidc_client_secret: Creating...
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
vault_kv_secret_v2.study_preview_keycloak_url: Creation complete after 0s [id=secret/data/secretspec/study/preview/KEYCLOAK_URL]
keycloak_openid_group_membership_protocol_mapper.study_staging: Creating...
vault_kv_secret_v2.study_dev_project_group: Creation complete after 0s [id=secret/data/secretspec/study/dev/PROJECT_GROUP]
vault_kv_secret_v2.study_preview_oidc_client_secret: Creating...
vault_kv_secret_v2.study_preview_project_group: Creation complete after 0s [id=secret/data/secretspec/study/preview/PROJECT_GROUP]
vault_kv_secret_v2.study_staging_oidc_client_id: Creating...
vault_kv_secret_v2.study_dev_oidc_client_secret: Creation complete after 0s [id=secret/data/secretspec/study/dev/OIDC_CLIENT_SECRET]
vault_kv_secret_v2.study_prod_oidc_client_secret: Creation complete after 0s [id=secret/data/secretspec/study/prod/OIDC_CLIENT_SECRET]
vault_kv_secret_v2.study_dev_oidc_client_id: Creation complete after 0s [id=secret/data/secretspec/study/dev/OIDC_CLIENT_ID]
vault_kv_secret_v2.study_staging_oidc_client_secret: Creating...
vault_kv_secret_v2.study_preview_oidc_client_id: Creating...
keycloak_openid_group_membership_protocol_mapper.study: Creating...
keycloak_group.project_cmu-study_study_admins: Creation complete after 0s [id=a683866f-c161-4298-8d00-911d611e2895]
keycloak_openid_group_membership_protocol_mapper.study_dev: Creation complete after 0s [id=d597e770-36f4-43bf-b9e3-04fa5228b29a]
vault_kv_secret_v2.study_prod_oidc_client_id: Creating...
keycloak_group_memberships.project_cmu-study_study_admins: Creating...
vault_kv_secret_v2.study_staging_oidc_client_id: Creation complete after 0s [id=secret/data/secretspec/study/staging/OIDC_CLIENT_ID]
vault_kv_secret_v2.study_preview_oidc_client_secret: Creation complete after 0s [id=secret/data/secretspec/study/preview/OIDC_CLIENT_SECRET]
keycloak_openid_group_membership_protocol_mapper.study_staging: Creation complete after 0s [id=7c989eb0-3f73-4692-b66c-4712c956f4ca]
vault_kv_secret_v2.study_staging_oidc_client_secret: Creation complete after 0s [id=secret/data/secretspec/study/staging/OIDC_CLIENT_SECRET]
vault_kv_secret_v2.study_preview_oidc_client_id: Creation complete after 0s [id=secret/data/secretspec/study/preview/OIDC_CLIENT_ID]
keycloak_openid_group_membership_protocol_mapper.study: Creation complete after 0s [id=54904eb2-e8ce-4ed7-9e50-565f4314ceeb]
vault_kv_secret_v2.study_prod_oidc_client_id: Creation complete after 0s [id=secret/data/secretspec/study/prod/OIDC_CLIENT_ID]
keycloak_group_memberships.project_cmu-study_study: Creation complete after 1s [id=scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a]
keycloak_group_memberships.project_cmu-study_study_admins: Creation complete after 1s [id=scottylabs/group-memberships/a683866f-c161-4298-8d00-911d611e2895]
keycloak_group_memberships.projects: Modifications complete after 2s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 38 added, 1 changed, 0 destroyed.

8. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.47s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_role.cmu-study: Creating...
discord_role.cmu-study: Creation complete after 1s [id=1550659990266581074]
discord_member_roles.annadavi: Creating...
discord_channel_permission.cmu-study_ch0: Creating...
discord_channel_permission.cmu-study_ch0: Creation complete after 0s [id=1512270354943578132:1550659990266581074:role]
discord_member_roles.annadavi: Creation complete after 3s [id=1353118209024331948:991766392590311455]

Apply complete! Resources: 3 added, 0 changed, 0 destroyed.

9. project: matrix_bridges dir: tofu/matrix_bridges workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.44s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_bridge_login_cmu_study: Creating...
null_resource.slack_bridge_login_cmu_study: Provisioning with 'local-exec'...
null_resource.slack_bridge_login_cmu_study (local-exec): Executing: ["/bin/sh" "-c" "governance slack-join --channel C09KN6M3P53"]
null_resource.slack_bridge_login_cmu_study: Creation complete after 0s [id=8035822658338344181]
synapse_mautrix_slack_link.cmu_study: Creating...
synapse_mautrix_slack_link.cmu_study: Creation complete after 9s [id=1512270354943578132]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

10. project: git dir: tofu/git workspace: default

Apply Error

Show Output
running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu apply -input=false /var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git/git-default.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git': exit status 1
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.30s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team.study_leads: Creating...
forgejo_team.cmu-study: Creating...
forgejo_repository.cmu-study_study: Creating...
forgejo_team.study_leads: Creation complete after 0s [name=study-leads]
forgejo_team_member.study_leads_annadavi: Creating...
forgejo_team.cmu-study: Creation complete after 0s [name=cmu-study]
forgejo_team_member.cmu-study_annadavi: Creating...
forgejo_team_member.study_leads_annadavi: Creation complete after 0s
github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Destroying... [id=dining-api:161772815]
github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Destroying... [id=eats:161775995]
github_team.cmu-study: Creating...
github_repository.cmu-study_study: Creating...
github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Destruction complete after 0s
forgejo_repository.cmu-eats_eats: Modifying... [name=eats]
forgejo_repository.cmu-eats_eats: Modifications complete after 0s [name=eats]
restapi_object.cmu-eats_eats_push_mirror: Creating...
restapi_object.cmu-eats_eats_kennel_webhook: Creating...
forgejo_team_repository.cmu-eats_eats: Creating...
forgejo_team_repository.cmu-eats_eats: Creation complete after 1s
forgejo_team_repository.eats_leads_eats: Creating...
restapi_object.cmu-eats_eats_push_mirror: Creation complete after 1s [id=remote_mirror_j6fIO7zNJdP]
restapi_object.cmu-eats_eats_kennel_webhook: Creation complete after 1s [id=32]
github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Creating...
forgejo_team_repository.eats_leads_eats: Creation complete after 0s
github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Destruction complete after 2s
forgejo_repository.cmu-eats_dining_api: Modifying... [name=dining-api]
forgejo_repository.cmu-eats_dining_api: Modifications complete after 0s [name=dining-api]
forgejo_team_repository.cmu-eats_dining_api: Creating...
restapi_object.cmu-eats_dining_api_push_mirror: Creating...
forgejo_team_repository.cmu-eats_dining_api: Creation complete after 0s
forgejo_team_repository.eats_leads_dining_api: Creating...
restapi_object.cmu-eats_dining_api_push_mirror: Creation complete after 0s [id=remote_mirror_6NT90ezmhct]
github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Creating...
forgejo_team_repository.eats_leads_dining_api: Creation complete after 0s
github_team.cmu-study: Still creating... [10s elapsed]
github_repository.cmu-study_study: Still creating... [10s elapsed]
github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Still creating... [10s elapsed]
github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Creation complete after 10s [id=eats:163759531]
github_repository_ruleset.cmu-eats_eats_ruleset: Modifying... [id=21330218]
github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Creation complete after 10s [id=dining-api:163759533]
github_repository_ruleset.cmu-eats_dining_api_ruleset: Modifying... [id=21879626]
github_team.cmu-study: Creation complete after 12s [id=19579842]
github_team_membership.cmu-study_annadavi: Creating...
github_repository_ruleset.cmu-eats_eats_ruleset: Modifications complete after 6s [id=21330218]
github_repository_ruleset.cmu-eats_dining_api_ruleset: Modifications complete after 6s [id=21879626]
github_team_membership.cmu-study_annadavi: Creation complete after 8s [id=19579842:annadavid1]
github_repository.cmu-study_study: Creation complete after 20s [id=study]
github_actions_repository_permissions.cmu-study_study_actions: Creating...
github_actions_repository_permissions.cmu-study_study_actions: Creation complete after 1s [id=study]
╷
│ Error: Unable to create team member
│ 
│   with forgejo_team_member.cmu-study_annadavi,
│   on forgejo_memberships.tf.json line 151, in resource.forgejo_team_member.cmu-study_annadavi:
│  151:       },
│ 
│ Unknown error (status 500): 
╵
╷
│ Error: Unable to create repository
│ 
│   with forgejo_repository.cmu-study_study,
│   on forgejo_repos.tf.json line 185, in resource.forgejo_repository.cmu-study_study:
│  185:       },
│ 
│ Repository with name "study" already exists: The repository with the same
│ name already exists.
╵


Apply Summary

10 projects, 9 successful, 0 failed, 1 errored

Ran Apply for 10 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `openbao` dir: `tofu/openbao` workspace: `default` 1. project: `posthog` dir: `tofu/posthog` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `sentry` dir: `tofu/sentry` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.38s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.26s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_cmu-study_ch0_annadavi: Creating... null_resource.slack_leads_annadavi: Creating... null_resource.slack_hub_annadavi: Creating... null_resource.slack_cmu-study_ch0_annadavi: Provisioning with 'local-exec'... null_resource.slack_cmu-study_ch0_annadavi (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C09KN6M3P53 --user U07M24AKRGV"] null_resource.slack_leads_annadavi: Provisioning with 'local-exec'... null_resource.slack_leads_annadavi (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0AN1M0HDV0 --user U07M24AKRGV"] null_resource.slack_hub_annadavi: Provisioning with 'local-exec'... null_resource.slack_hub_annadavi (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U07M24AKRGV"] null_resource.slack_hub_annadavi: Creation complete after 1s [id=968070886944939043] null_resource.slack_cmu-study_ch0_annadavi: Creation complete after 1s [id=180323051751415711] null_resource.slack_leads_annadavi: Creation complete after 1s [id=4222181416230461163] Apply complete! Resources: 3 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `openbao` dir: `tofu/openbao` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.29s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ vault_identity_group.study_admins: Creating... vault_identity_group.cmugpt-surface_admins: Modifying... [id=3ae18ef3-f370-f163-767a-accfb092ea0d] vault_identity_group.discord-verify_admins: Modifying... [id=c3e29b3b-0ce5-61a3-3a69-e950f297fa78] vault_identity_group.changelog_admins: Modifying... [id=d73c0ef4-a7d2-d171-ebd1-1c15cee174c8] vault_identity_group.eats_admins: Modifying... [id=05959815-c896-9703-32eb-7ba84604303c] vault_identity_group.cal_admins: Modifying... [id=da96ea82-8537-f0a1-0d42-a7ab1cc3c006] vault_identity_group.components_admins: Modifying... [id=d2971366-64a4-984e-1e4e-dad3cd07c841] vault_identity_group.housing_admins: Modifying... [id=7c688f49-59f1-1593-1e19-16f70b86eae2] vault_identity_group.insta_admins: Modifying... [id=be24e607-e8d5-9fa3-a11a-9a8ca05ede04] vault_identity_group.scottylabs-org_admins: Modifying... [id=0dbd914d-b96c-de7b-1ec5-a0ef330b20e0] vault_identity_group.cal_admins: Modifications complete after 0s [id=da96ea82-8537-f0a1-0d42-a7ab1cc3c006] vault_identity_group.cmugpt-surface_admins: Modifications complete after 0s [id=3ae18ef3-f370-f163-767a-accfb092ea0d] vault_identity_group.scottylabs-org_admins: Modifications complete after 0s [id=0dbd914d-b96c-de7b-1ec5-a0ef330b20e0] vault_identity_group.changelog_admins: Modifications complete after 0s [id=d73c0ef4-a7d2-d171-ebd1-1c15cee174c8] vault_identity_group.study_admins: Creation complete after 0s [id=14e88827-391c-ab7b-acfa-b824a25f0c4a] vault_identity_group.discord-verify_admins: Modifications complete after 0s [id=c3e29b3b-0ce5-61a3-3a69-e950f297fa78] vault_identity_group.housing_admins: Modifications complete after 0s [id=7c688f49-59f1-1593-1e19-16f70b86eae2] vault_identity_group.components_admins: Modifications complete after 0s [id=d2971366-64a4-984e-1e4e-dad3cd07c841] vault_identity_group.occ-main_admins: Modifying... [id=4d40db70-1bac-505e-ec8d-5a9d5cf15ac7] vault_identity_group.insta_admins: Modifications complete after 0s [id=be24e607-e8d5-9fa3-a11a-9a8ca05ede04] vault_identity_group.eats_admins: Modifications complete after 0s [id=05959815-c896-9703-32eb-7ba84604303c] vault_policy.study_prod: Creating... vault_identity_group.study_members: Creating... vault_identity_group.cmugpt-agent_admins: Modifying... [id=cb081cd2-a3cd-81e1-8c20-0914cac5c213] vault_identity_group.maps_admins: Modifying... [id=6b2bd85e-e311-ae83-8a07-30100f0ac12f] vault_policy.study_dev: Creating... vault_identity_group.lost-and-found-v2_admins: Modifying... [id=db8b5072-bb86-c9d0-e93f-951c2901bd15] vault_identity_group.cmugpt-mcp-server_admins: Modifying... [id=a4717a6f-015f-9fae-c6a2-76657b8a7464] vault_identity_group.quest_admins: Modifying... [id=c03f2eab-1701-f315-fc17-39814ddcb34b] vault_identity_group.groupme-mirror_admins: Modifying... [id=e264aabf-8766-67fd-f0bf-860a240ee2c2] vault_identity_group.occ-main_admins: Modifications complete after 1s [id=4d40db70-1bac-505e-ec8d-5a9d5cf15ac7] vault_identity_group.documentation_admins: Modifying... [id=35c5df83-0e17-31fc-7b82-3365d5526dd5] vault_identity_group.maps_admins: Modifications complete after 1s [id=6b2bd85e-e311-ae83-8a07-30100f0ac12f] vault_identity_group.courses_admins: Modifying... [id=d090ae17-1992-86a5-1f8b-092c937092e8] vault_identity_group.lost-and-found-v2_admins: Modifications complete after 1s [id=db8b5072-bb86-c9d0-e93f-951c2901bd15] vault_identity_group.study_members: Creation complete after 1s [id=d9c27753-d898-8113-e4cd-1b3fb4b8f73b] vault_identity_group.cmugpt-agent_admins: Modifications complete after 1s [id=cb081cd2-a3cd-81e1-8c20-0914cac5c213] vault_identity_group.link-shortener_admins: Modifying... [id=d3fa8791-4bd4-db20-68b6-79db5ac36a25] vault_identity_group.bus-sign_admins: Modifying... [id=1b72d8b5-8511-fe92-c98c-2ddb8cbf17a1] vault_identity_group.cmugpt-mcp-server_admins: Modifications complete after 1s [id=a4717a6f-015f-9fae-c6a2-76657b8a7464] vault_identity_group.quest_admins: Modifications complete after 0s [id=c03f2eab-1701-f315-fc17-39814ddcb34b] vault_identity_group.tartan-vote_admins: Modifying... [id=ff77ad01-20b1-c320-37bd-6185b3f98d80] vault_identity_group.groupme-mirror_admins: Modifications complete after 0s [id=e264aabf-8766-67fd-f0bf-860a240ee2c2] vault_identity_group.terrier_admins: Modifying... [id=956692a1-84f6-c95c-6d40-2382ead13476] vault_identity_group.dalmatian_admins: Modifying... [id=b1d88b18-8b44-c2ea-9220-fc53ee7e3c7e] vault_identity_group.cmugpt-sms-surface_admins: Modifying... [id=780f9498-dee7-8cac-9bbc-e1feab5ea2fb] vault_policy.study_dev: Creation complete after 1s [id=study-dev] vault_policy.study_prod: Creation complete after 1s [id=study-prod] vault_identity_group.documentation_admins: Modifications complete after 0s [id=35c5df83-0e17-31fc-7b82-3365d5526dd5] vault_identity_group_alias.study_admins: Creating... vault_identity_group_alias.study_members: Creating... vault_identity_group.courses_admins: Modifications complete after 0s [id=d090ae17-1992-86a5-1f8b-092c937092e8] vault_identity_group.link-shortener_admins: Modifications complete after 0s [id=d3fa8791-4bd4-db20-68b6-79db5ac36a25] vault_identity_group.bus-sign_admins: Modifications complete after 0s [id=1b72d8b5-8511-fe92-c98c-2ddb8cbf17a1] vault_identity_group.tartan-vote_admins: Modifications complete after 0s [id=ff77ad01-20b1-c320-37bd-6185b3f98d80] vault_identity_group.cmugpt-sms-surface_admins: Modifications complete after 0s [id=780f9498-dee7-8cac-9bbc-e1feab5ea2fb] vault_identity_group.dalmatian_admins: Modifications complete after 0s [id=b1d88b18-8b44-c2ea-9220-fc53ee7e3c7e] vault_identity_group.terrier_admins: Modifications complete after 0s [id=956692a1-84f6-c95c-6d40-2382ead13476] vault_identity_group_alias.study_admins: Creation complete after 0s [id=f515d97d-daff-e0c3-f762-00a95eb26ea3] vault_identity_group_alias.study_members: Creation complete after 0s [id=4426b35a-0210-225b-1afb-b79089f12a41] Apply complete! Resources: 6 added, 24 changed, 0 destroyed. ``` </details> --- ### 4. project: `posthog` dir: `tofu/posthog` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.47s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ posthog_project.cmu-study_study: Creating... vault_kv_secret_v2.cmu-study_study_posthog_host: Creating... vault_kv_secret_v2.cmu-study_study_posthog_host: Creation complete after 0s [id=secret/data/secretspec/study/prod/POSTHOG_HOST] posthog_project.cmu-study_study: Creation complete after 1s [name=study] vault_kv_secret_v2.cmu-study_study_posthog_key: Creating... vault_kv_secret_v2.cmu-study_study_posthog_key: Creation complete after 0s [id=secret/data/secretspec/study/prod/POSTHOG_KEY] Apply complete! Resources: 3 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.53s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_annadavi: Creating... google_cloud_identity_group_membership.ops_annadavi: Creating... google_cloud_identity_group_membership.ops_annadavi: Creation complete after 1s [id=groups/03fwokq036zsp1f/memberships/113975979101831473222] google_cloud_identity_group_membership.tech_annadavi: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/113975979101831473222] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 6. project: `sentry` dir: `tofu/sentry` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.37s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ sentry_team.cmu-study: Creating... sentry_team.cmu-study: Creation complete after 0s [id=cmu-study] sentry_project.cmu-study_study: Creating... sentry_project.cmu-study_study: Creation complete after 2s [id=study] sentry_key.cmu-study_study: Creating... sentry_key.cmu-study_study: Creation complete after 0s [id=e1d434c547505c7213a3b2618030187c] vault_kv_secret_v2.cmu-study_study_sentry_dsn: Creating... vault_kv_secret_v2.cmu-study_study_sentry_dsn: Creation complete after 1s [id=secret/data/secretspec/study/prod/SENTRY_DSN] Apply complete! Resources: 4 added, 0 changed, 0 destroyed. ``` </details> --- ### 7. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.39s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group.project_cmu-study_study: Creating... keycloak_openid_client.study_dev: Creating... keycloak_openid_client.study_staging: Creating... keycloak_openid_client.study: Creating... vault_kv_secret_v2.study_preview_keycloak_realm: Creating... vault_kv_secret_v2.study_staging_keycloak_url: Creating... vault_kv_secret_v2.study_staging_project_admin_group: Creating... vault_kv_secret_v2.study_dev_project_admin_group: Creating... vault_kv_secret_v2.study_dev_oauth_relay_url: Creating... vault_kv_secret_v2.study_staging_project_group: Creating... keycloak_group.project_cmu-study_study: Creation complete after 0s [id=f1b424fe-27f7-40ef-831f-9885f66bac3a] vault_kv_secret_v2.study_prod_keycloak_realm: Creating... keycloak_openid_client.study_dev: Creation complete after 0s [id=1983432b-c99e-4470-9a59-1f8fdb9f8bef] vault_kv_secret_v2.study_prod_oauth_relay_url: Creating... keycloak_openid_client.study_staging: Creation complete after 0s [id=18b84a0e-d9fa-4e99-9456-db6fba3358ff] keycloak_openid_client.study: Creation complete after 0s [id=7622bc88-d377-4e9b-8995-86ec8e6f4638] vault_kv_secret_v2.study_prod_keycloak_url: Creating... vault_kv_secret_v2.study_preview_project_admin_group: Creating... vault_kv_secret_v2.study_dev_project_admin_group: Creation complete after 0s [id=secret/data/secretspec/study/dev/PROJECT_ADMIN_GROUP] vault_kv_secret_v2.study_staging_project_group: Creation complete after 0s [id=secret/data/secretspec/study/staging/PROJECT_GROUP] vault_kv_secret_v2.study_staging_project_admin_group: Creation complete after 0s [id=secret/data/secretspec/study/staging/PROJECT_ADMIN_GROUP] vault_kv_secret_v2.study_dev_oauth_relay_url: Creation complete after 0s [id=secret/data/secretspec/study/dev/OAUTH_RELAY_URL] vault_kv_secret_v2.study_staging_keycloak_url: Creation complete after 0s [id=secret/data/secretspec/study/staging/KEYCLOAK_URL] vault_kv_secret_v2.study_preview_keycloak_realm: Creation complete after 0s [id=secret/data/secretspec/study/preview/KEYCLOAK_REALM] vault_kv_secret_v2.study_prod_keycloak_realm: Creation complete after 0s [id=secret/data/secretspec/study/prod/KEYCLOAK_REALM] vault_kv_secret_v2.study_staging_keycloak_realm: Creating... vault_kv_secret_v2.study_dev_keycloak_url: Creating... vault_kv_secret_v2.study_dev_keycloak_realm: Creating... vault_kv_secret_v2.study_staging_oauth_relay_url: Creating... vault_kv_secret_v2.study_prod_project_group: Creating... vault_kv_secret_v2.study_prod_project_admin_group: Creating... vault_kv_secret_v2.study_preview_oauth_relay_url: Creating... vault_kv_secret_v2.study_prod_keycloak_url: Creation complete after 1s [id=secret/data/secretspec/study/prod/KEYCLOAK_URL] vault_kv_secret_v2.study_preview_keycloak_url: Creating... vault_kv_secret_v2.study_prod_oauth_relay_url: Creation complete after 1s [id=secret/data/secretspec/study/prod/OAUTH_RELAY_URL] vault_kv_secret_v2.study_preview_project_group: Creating... vault_kv_secret_v2.study_preview_project_admin_group: Creation complete after 1s [id=secret/data/secretspec/study/preview/PROJECT_ADMIN_GROUP] vault_kv_secret_v2.study_dev_project_group: Creating... vault_kv_secret_v2.study_staging_oauth_relay_url: Creation complete after 1s [id=secret/data/secretspec/study/staging/OAUTH_RELAY_URL] vault_kv_secret_v2.study_staging_keycloak_realm: Creation complete after 1s [id=secret/data/secretspec/study/staging/KEYCLOAK_REALM] vault_kv_secret_v2.study_preview_oauth_relay_url: Creation complete after 1s [id=secret/data/secretspec/study/preview/OAUTH_RELAY_URL] vault_kv_secret_v2.study_dev_keycloak_url: Creation complete after 1s [id=secret/data/secretspec/study/dev/KEYCLOAK_URL] vault_kv_secret_v2.study_dev_keycloak_realm: Creation complete after 1s [id=secret/data/secretspec/study/dev/KEYCLOAK_REALM] keycloak_openid_group_membership_protocol_mapper.study_dev: Creating... keycloak_group.project_cmu-study_study_admins: Creating... vault_kv_secret_v2.study_prod_project_admin_group: Creation complete after 1s [id=secret/data/secretspec/study/prod/PROJECT_ADMIN_GROUP] keycloak_group_memberships.project_cmu-study_study: Creating... vault_kv_secret_v2.study_prod_project_group: Creation complete after 1s [id=secret/data/secretspec/study/prod/PROJECT_GROUP] vault_kv_secret_v2.study_dev_oidc_client_secret: Creating... vault_kv_secret_v2.study_dev_oidc_client_id: Creating... vault_kv_secret_v2.study_prod_oidc_client_secret: Creating... keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] vault_kv_secret_v2.study_preview_keycloak_url: Creation complete after 0s [id=secret/data/secretspec/study/preview/KEYCLOAK_URL] keycloak_openid_group_membership_protocol_mapper.study_staging: Creating... vault_kv_secret_v2.study_dev_project_group: Creation complete after 0s [id=secret/data/secretspec/study/dev/PROJECT_GROUP] vault_kv_secret_v2.study_preview_oidc_client_secret: Creating... vault_kv_secret_v2.study_preview_project_group: Creation complete after 0s [id=secret/data/secretspec/study/preview/PROJECT_GROUP] vault_kv_secret_v2.study_staging_oidc_client_id: Creating... vault_kv_secret_v2.study_dev_oidc_client_secret: Creation complete after 0s [id=secret/data/secretspec/study/dev/OIDC_CLIENT_SECRET] vault_kv_secret_v2.study_prod_oidc_client_secret: Creation complete after 0s [id=secret/data/secretspec/study/prod/OIDC_CLIENT_SECRET] vault_kv_secret_v2.study_dev_oidc_client_id: Creation complete after 0s [id=secret/data/secretspec/study/dev/OIDC_CLIENT_ID] vault_kv_secret_v2.study_staging_oidc_client_secret: Creating... vault_kv_secret_v2.study_preview_oidc_client_id: Creating... keycloak_openid_group_membership_protocol_mapper.study: Creating... keycloak_group.project_cmu-study_study_admins: Creation complete after 0s [id=a683866f-c161-4298-8d00-911d611e2895] keycloak_openid_group_membership_protocol_mapper.study_dev: Creation complete after 0s [id=d597e770-36f4-43bf-b9e3-04fa5228b29a] vault_kv_secret_v2.study_prod_oidc_client_id: Creating... keycloak_group_memberships.project_cmu-study_study_admins: Creating... vault_kv_secret_v2.study_staging_oidc_client_id: Creation complete after 0s [id=secret/data/secretspec/study/staging/OIDC_CLIENT_ID] vault_kv_secret_v2.study_preview_oidc_client_secret: Creation complete after 0s [id=secret/data/secretspec/study/preview/OIDC_CLIENT_SECRET] keycloak_openid_group_membership_protocol_mapper.study_staging: Creation complete after 0s [id=7c989eb0-3f73-4692-b66c-4712c956f4ca] vault_kv_secret_v2.study_staging_oidc_client_secret: Creation complete after 0s [id=secret/data/secretspec/study/staging/OIDC_CLIENT_SECRET] vault_kv_secret_v2.study_preview_oidc_client_id: Creation complete after 0s [id=secret/data/secretspec/study/preview/OIDC_CLIENT_ID] keycloak_openid_group_membership_protocol_mapper.study: Creation complete after 0s [id=54904eb2-e8ce-4ed7-9e50-565f4314ceeb] vault_kv_secret_v2.study_prod_oidc_client_id: Creation complete after 0s [id=secret/data/secretspec/study/prod/OIDC_CLIENT_ID] keycloak_group_memberships.project_cmu-study_study: Creation complete after 1s [id=scottylabs/group-memberships/f1b424fe-27f7-40ef-831f-9885f66bac3a] keycloak_group_memberships.project_cmu-study_study_admins: Creation complete after 1s [id=scottylabs/group-memberships/a683866f-c161-4298-8d00-911d611e2895] keycloak_group_memberships.projects: Modifications complete after 2s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 38 added, 1 changed, 0 destroyed. ``` </details> --- ### 8. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.47s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_role.cmu-study: Creating... discord_role.cmu-study: Creation complete after 1s [id=1550659990266581074] discord_member_roles.annadavi: Creating... discord_channel_permission.cmu-study_ch0: Creating... discord_channel_permission.cmu-study_ch0: Creation complete after 0s [id=1512270354943578132:1550659990266581074:role] discord_member_roles.annadavi: Creation complete after 3s [id=1353118209024331948:991766392590311455] Apply complete! Resources: 3 added, 0 changed, 0 destroyed. ``` </details> --- ### 9. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.44s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_bridge_login_cmu_study: Creating... null_resource.slack_bridge_login_cmu_study: Provisioning with 'local-exec'... null_resource.slack_bridge_login_cmu_study (local-exec): Executing: ["/bin/sh" "-c" "governance slack-join --channel C09KN6M3P53"] null_resource.slack_bridge_login_cmu_study: Creation complete after 0s [id=8035822658338344181] synapse_mautrix_slack_link.cmu_study: Creating... synapse_mautrix_slack_link.cmu_study: Creation complete after 9s [id=1512270354943578132] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 10. project: `git` dir: `tofu/git` workspace: `default` **Apply Error** <details><summary>Show Output</summary> ``` running '/nix/store/6knnwrwxlzs7f7rlr14b8agk69aa3w5b-opentofu-1.12.6/bin/tofu apply -input=false /var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git/git-default.tfplan' in '/var/lib/atlantis/repos/ScottyLabs/governance/215/default/tofu/git': exit status 1 Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.30s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team.study_leads: Creating... forgejo_team.cmu-study: Creating... forgejo_repository.cmu-study_study: Creating... forgejo_team.study_leads: Creation complete after 0s [name=study-leads] forgejo_team_member.study_leads_annadavi: Creating... forgejo_team.cmu-study: Creation complete after 0s [name=cmu-study] forgejo_team_member.cmu-study_annadavi: Creating... forgejo_team_member.study_leads_annadavi: Creation complete after 0s github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Destroying... [id=dining-api:161772815] github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Destroying... [id=eats:161775995] github_team.cmu-study: Creating... github_repository.cmu-study_study: Creating... github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Destruction complete after 0s forgejo_repository.cmu-eats_eats: Modifying... [name=eats] forgejo_repository.cmu-eats_eats: Modifications complete after 0s [name=eats] restapi_object.cmu-eats_eats_push_mirror: Creating... restapi_object.cmu-eats_eats_kennel_webhook: Creating... forgejo_team_repository.cmu-eats_eats: Creating... forgejo_team_repository.cmu-eats_eats: Creation complete after 1s forgejo_team_repository.eats_leads_eats: Creating... restapi_object.cmu-eats_eats_push_mirror: Creation complete after 1s [id=remote_mirror_j6fIO7zNJdP] restapi_object.cmu-eats_eats_kennel_webhook: Creation complete after 1s [id=32] github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Creating... forgejo_team_repository.eats_leads_eats: Creation complete after 0s github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Destruction complete after 2s forgejo_repository.cmu-eats_dining_api: Modifying... [name=dining-api] forgejo_repository.cmu-eats_dining_api: Modifications complete after 0s [name=dining-api] forgejo_team_repository.cmu-eats_dining_api: Creating... restapi_object.cmu-eats_dining_api_push_mirror: Creating... forgejo_team_repository.cmu-eats_dining_api: Creation complete after 0s forgejo_team_repository.eats_leads_dining_api: Creating... restapi_object.cmu-eats_dining_api_push_mirror: Creation complete after 0s [id=remote_mirror_6NT90ezmhct] github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Creating... forgejo_team_repository.eats_leads_dining_api: Creation complete after 0s github_team.cmu-study: Still creating... [10s elapsed] github_repository.cmu-study_study: Still creating... [10s elapsed] github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Still creating... [10s elapsed] github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Creation complete after 10s [id=eats:163759531] github_repository_ruleset.cmu-eats_eats_ruleset: Modifying... [id=21330218] github_repository_deploy_key.cmu-eats_dining_api_mirror_deploy_key: Creation complete after 10s [id=dining-api:163759533] github_repository_ruleset.cmu-eats_dining_api_ruleset: Modifying... [id=21879626] github_team.cmu-study: Creation complete after 12s [id=19579842] github_team_membership.cmu-study_annadavi: Creating... github_repository_ruleset.cmu-eats_eats_ruleset: Modifications complete after 6s [id=21330218] github_repository_ruleset.cmu-eats_dining_api_ruleset: Modifications complete after 6s [id=21879626] github_team_membership.cmu-study_annadavi: Creation complete after 8s [id=19579842:annadavid1] github_repository.cmu-study_study: Creation complete after 20s [id=study] github_actions_repository_permissions.cmu-study_study_actions: Creating... github_actions_repository_permissions.cmu-study_study_actions: Creation complete after 1s [id=study] ╷ │ Error: Unable to create team member │ │ with forgejo_team_member.cmu-study_annadavi, │ on forgejo_memberships.tf.json line 151, in resource.forgejo_team_member.cmu-study_annadavi: │ 151: }, │ │ Unknown error (status 500): ╵ ╷ │ Error: Unable to create repository │ │ with forgejo_repository.cmu-study_study, │ on forgejo_repos.tf.json line 185, in resource.forgejo_repository.cmu-study_study: │ 185: }, │ │ Repository with name "study" already exists: The repository with the same │ name already exists. ╵ ``` </details> --- ### Apply Summary 10 projects, 9 successful, 0 failed, 1 errored
Owner

atlantis import -p git forgejo_repository.cmu-study_study ScottyLabs/study

atlantis import -p git forgejo_repository.cmu-study_study ScottyLabs/study

Ran Import for project: git dir: tofu/git workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.06s

data.external.identity_rhsen: Reading...
data.external.identity_dkou: Reading...
data.external.identity_kluoxx: Reading...
data.external.identity_krishsax: Reading...
data.external.identity_siddharthr: Reading...
data.external.identity_kritdass: Reading...
data.external.identity_anish: Reading...
data.external.identity_emmawong: Reading...
data.external.identity_jc_965: Reading...
data.external.identity_kluoxx: Read complete after 1s [id=-]
data.external.identity_stringatlas: Reading...
data.external.identity_siddharthr: Read complete after 1s [id=-]
data.external.identity_etashjha: Reading...
data.external.identity_krishsax: Read complete after 1s [id=-]
data.external.identity_jw8: Reading...
data.external.identity_dkou: Read complete after 1s [id=-]
data.external.identity_tentype: Reading...
data.external.identity_anish: Read complete after 1s [id=-]
data.external.identity_kothurvidur: Reading...
data.external.identity_rhsen: Read complete after 1s [id=-]
data.external.identity_maybe_yiyi: Reading...
data.external.identity_emmawong: Read complete after 1s [id=-]
data.external.identity_gabrielx: Reading...
data.external.identity_kritdass: Read complete after 1s [id=-]
data.external.identity_jalenluorion: Reading...
data.external.identity_jc_965: Read complete after 0s [id=-]
data.external.identity_amzoeee: Reading...
data.external.identity_thesuperrl: Reading...
data.external.identity_etashjha: Read complete after 0s [id=-]
data.external.identity_zoexue: Reading...
data.external.identity_jw8: Read complete after 0s [id=-]
data.external.identity_liamwalk: Reading...
data.external.identity_stringatlas: Read complete after 0s [id=-]
data.external.identity_qianxuege: Reading...
data.external.identity_tentype: Read complete after 0s [id=-]
data.external.identity_jessica_lr_song: Reading...
data.external.identity_kothurvidur: Read complete after 0s [id=-]
data.external.identity_dcrodrig: Reading...
data.external.identity_gabrielx: Read complete after 0s [id=-]
data.external.identity_annadavi: Reading...
data.external.identity_jalenluorion: Read complete after 0s [id=-]
data.external.identity_rzhu3: Reading...
data.external.identity_maybe_yiyi: Read complete after 0s [id=-]
data.external.identity_brightz: Reading...
data.external.identity_amzoeee: Read complete after 0s [id=-]
data.external.identity_jiyas: Reading...
data.external.identity_thesuperrl: Read complete after 0s [id=-]
data.external.identity_manrajs: Reading...
data.external.identity_liamwalk: Read complete after 0s [id=-]
data.external.identity_xboxbedrock: Reading...
data.external.identity_qianxuege: Read complete after 0s [id=-]
data.external.identity_riztofu: Reading...
data.external.identity_dcrodrig: Read complete after 0s [id=-]
data.external.identity_annadavi: Read complete after 0s [id=-]
data.external.identity_gostmeaper: Reading...
data.external.identity_nikhilr: Reading...
data.external.identity_jessica_lr_song: Read complete after 0s [id=-]
data.external.identity_acchs: Reading...
data.external.identity_zoexue: Read complete after 0s [id=-]
data.external.identity_arom1a: Reading...
data.external.identity_brightz: Read complete after 0s [id=-]
data.external.identity_kechezon: Reading...
data.external.identity_rzhu3: Read complete after 0s [id=-]
data.forgejo_organization.this: Reading...
data.external.identity_jiyas: Read complete after 0s [id=-]
forgejo_repository.cmu-study_study: Importing from ID "ScottyLabs/study"...
data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs]
forgejo_repository.cmu-study_study: Import prepared!
  Prepared forgejo_repository for import
data.external.identity_manrajs: Read complete after 0s [id=-]
data.external.identity_xboxbedrock: Read complete after 0s [id=-]
data.external.identity_nikhilr: Read complete after 0s [id=-]
data.external.identity_gostmeaper: Read complete after 0s [id=-]
data.external.identity_acchs: Read complete after 0s [id=-]
data.external.identity_arom1a: Read complete after 0s [id=-]
data.external.identity_riztofu: Read complete after 0s [id=-]
forgejo_repository.cmu-study_study: Refreshing state... [name=study]
data.external.identity_kechezon: Read complete after 0s [id=-]

Import successful!

The resources that were imported are shown above. These resources are now in
your OpenTofu state and will henceforth be managed by OpenTofu.
🚮 A plan file was discarded. Re-plan would be required before applying.
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    
Ran Import for project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.06s data.external.identity_rhsen: Reading... data.external.identity_dkou: Reading... data.external.identity_kluoxx: Reading... data.external.identity_krishsax: Reading... data.external.identity_siddharthr: Reading... data.external.identity_kritdass: Reading... data.external.identity_anish: Reading... data.external.identity_emmawong: Reading... data.external.identity_jc_965: Reading... data.external.identity_kluoxx: Read complete after 1s [id=-] data.external.identity_stringatlas: Reading... data.external.identity_siddharthr: Read complete after 1s [id=-] data.external.identity_etashjha: Reading... data.external.identity_krishsax: Read complete after 1s [id=-] data.external.identity_jw8: Reading... data.external.identity_dkou: Read complete after 1s [id=-] data.external.identity_tentype: Reading... data.external.identity_anish: Read complete after 1s [id=-] data.external.identity_kothurvidur: Reading... data.external.identity_rhsen: Read complete after 1s [id=-] data.external.identity_maybe_yiyi: Reading... data.external.identity_emmawong: Read complete after 1s [id=-] data.external.identity_gabrielx: Reading... data.external.identity_kritdass: Read complete after 1s [id=-] data.external.identity_jalenluorion: Reading... data.external.identity_jc_965: Read complete after 0s [id=-] data.external.identity_amzoeee: Reading... data.external.identity_thesuperrl: Reading... data.external.identity_etashjha: Read complete after 0s [id=-] data.external.identity_zoexue: Reading... data.external.identity_jw8: Read complete after 0s [id=-] data.external.identity_liamwalk: Reading... data.external.identity_stringatlas: Read complete after 0s [id=-] data.external.identity_qianxuege: Reading... data.external.identity_tentype: Read complete after 0s [id=-] data.external.identity_jessica_lr_song: Reading... data.external.identity_kothurvidur: Read complete after 0s [id=-] data.external.identity_dcrodrig: Reading... data.external.identity_gabrielx: Read complete after 0s [id=-] data.external.identity_annadavi: Reading... data.external.identity_jalenluorion: Read complete after 0s [id=-] data.external.identity_rzhu3: Reading... data.external.identity_maybe_yiyi: Read complete after 0s [id=-] data.external.identity_brightz: Reading... data.external.identity_amzoeee: Read complete after 0s [id=-] data.external.identity_jiyas: Reading... data.external.identity_thesuperrl: Read complete after 0s [id=-] data.external.identity_manrajs: Reading... data.external.identity_liamwalk: Read complete after 0s [id=-] data.external.identity_xboxbedrock: Reading... data.external.identity_qianxuege: Read complete after 0s [id=-] data.external.identity_riztofu: Reading... data.external.identity_dcrodrig: Read complete after 0s [id=-] data.external.identity_annadavi: Read complete after 0s [id=-] data.external.identity_gostmeaper: Reading... data.external.identity_nikhilr: Reading... data.external.identity_jessica_lr_song: Read complete after 0s [id=-] data.external.identity_acchs: Reading... data.external.identity_zoexue: Read complete after 0s [id=-] data.external.identity_arom1a: Reading... data.external.identity_brightz: Read complete after 0s [id=-] data.external.identity_kechezon: Reading... data.external.identity_rzhu3: Read complete after 0s [id=-] data.forgejo_organization.this: Reading... data.external.identity_jiyas: Read complete after 0s [id=-] forgejo_repository.cmu-study_study: Importing from ID "ScottyLabs/study"... data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs] forgejo_repository.cmu-study_study: Import prepared! Prepared forgejo_repository for import data.external.identity_manrajs: Read complete after 0s [id=-] data.external.identity_xboxbedrock: Read complete after 0s [id=-] data.external.identity_nikhilr: Read complete after 0s [id=-] data.external.identity_gostmeaper: Read complete after 0s [id=-] data.external.identity_acchs: Read complete after 0s [id=-] data.external.identity_arom1a: Read complete after 0s [id=-] data.external.identity_riztofu: Read complete after 0s [id=-] forgejo_repository.cmu-study_study: Refreshing state... [name=study] data.external.identity_kechezon: Read complete after 0s [id=-] Import successful! The resources that were imported are shown above. These resources are now in your OpenTofu state and will henceforth be managed by OpenTofu. ``` </details> :put_litter_in_its_place: A plan file was discarded. Re-plan would be required before applying. * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ```
Owner

atlantis plan -p git

atlantis plan -p git

Ran Plan for project: git dir: tofu/git workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.05s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # forgejo_repository.cmu-study_study will be updated in-place
! resource "forgejo_repository" "cmu-study_study" {
!       clone_url                   = "https://git.cmu.dev/ScottyLabs/study.git" -> (known after apply)
!       created_at                  = "2026-09-18T22:52:34Z" -> (known after apply)
!       description                 = "Labrador Boilerplate for Web" -> "A platform for creating and joining study groups"
!       empty                       = false -> (known after apply)
+       external_tracker            = (known after apply)
+       external_wiki               = (known after apply)
!       forks_count                 = 0 -> (known after apply)
!       full_name                   = "ScottyLabs/study" -> (known after apply)
!       html_url                    = "https://git.cmu.dev/ScottyLabs/study" -> (known after apply)
        id                          = 93
!       internal_tracker            = {
!           allow_only_contributors_to_track_time = true -> (known after apply)
!           enable_issue_dependencies             = true -> (known after apply)
!           enable_time_tracker                   = true -> (known after apply)
        } -> (known after apply)
+       mirror_interval             = (known after apply)
!       mirror_updated              = "0001-01-01T00:00:00Z" -> (known after apply)
        name                        = "study"
!       open_issues_count           = 0 -> (known after apply)
!       open_pr_counter             = 0 -> (known after apply)
!       permissions                 = {
!           admin = true -> (known after apply)
!           pull  = true -> (known after apply)
!           push  = true -> (known after apply)
        } -> (known after apply)
!       release_counter             = 0 -> (known after apply)
!       size                        = 2234 -> (known after apply)
!       ssh_url                     = "ssh://forgejo@git.cmu.dev/ScottyLabs/study.git" -> (known after apply)
!       stars_count                 = 0 -> (known after apply)
!       updated_at                  = "2026-09-18T22:52:36Z" -> (known after apply)
!       watchers_count              = 3 -> (known after apply)
!       website                     = "https://study-group-finder-eight.vercel.app" -> "https://cmustudy.com"
        # (30 unchanged attributes hidden)
    }

  # forgejo_team_member.cmu-study_annadavi will be created
+ resource "forgejo_team_member" "cmu-study_annadavi" {
+       team_id = 43
+       user    = "annadavi"
    }

  # forgejo_team_repository.cmu-study_study will be created
+ resource "forgejo_team_repository" "cmu-study_study" {
+       owner      = "ScottyLabs"
+       repository = "study"
+       team_id    = 43
    }

  # forgejo_team_repository.study_leads_study will be created
+ resource "forgejo_team_repository" "study_leads_study" {
+       owner      = "ScottyLabs"
+       repository = "study"
+       team_id    = 42
    }

  # github_repository_deploy_key.cmu-study_study_mirror_deploy_key will be created
+ resource "github_repository_deploy_key" "cmu-study_study_mirror_deploy_key" {
+       etag       = (known after apply)
+       id         = (known after apply)
+       key        = (known after apply)
+       read_only  = false
+       repository = "study"
+       title      = "Forgejo mirror"
    }

  # github_repository_ruleset.cmu-study_study_ruleset will be created
+ resource "github_repository_ruleset" "cmu-study_study_ruleset" {
+       enforcement = "active"
+       etag        = (known after apply)
+       id          = (known after apply)
+       name        = "Default"
+       node_id     = (known after apply)
+       repository  = "study"
+       ruleset_id  = (known after apply)
+       target      = "branch"

+       bypass_actors {
+           actor_id    = (known after apply)
+           actor_type  = "DeployKey"
+           bypass_mode = "always"
        }

+       conditions {
+           ref_name {
+               exclude = []
+               include = [
+                   "~ALL",
                ]
            }
        }

+       rules {
+           creation                      = true
+           deletion                      = true
+           non_fast_forward              = true
+           update                        = true
+           update_allows_fetch_and_merge = false

+           pull_request {
+               allowed_merge_methods             = (known after apply)
+               dismiss_stale_reviews_on_push     = false
+               require_code_owner_review         = false
+               require_last_push_approval        = false
+               required_approving_review_count   = 1
+               required_review_thread_resolution = false
            }
        }
    }

  # restapi_object.cmu-study_study_kennel_webhook will be created
+ resource "restapi_object" "cmu-study_study_kennel_webhook" {
+       api_data        = (known after apply)
+       api_response    = (known after apply)
+       create_response = (known after apply)
+       data            = (sensitive value)
+       id              = (known after apply)
+       id_attribute    = "id"
+       path            = "/api/v1/repos/ScottyLabs/study/hooks"
+       update_method   = "PATCH"
    }

  # restapi_object.cmu-study_study_push_mirror will be created
+ resource "restapi_object" "cmu-study_study_push_mirror" {
+       api_data        = (known after apply)
+       api_response    = (known after apply)
+       create_response = (known after apply)
+       data            = jsonencode(
            {
+               interval       = "8h0m0s"
+               remote_address = "git@github.com:ScottyLabs/study.git"
+               sync_on_commit = true
+               use_ssh        = true
            }
        )
+       id              = (known after apply)
+       id_attribute    = "remote_name"
+       path            = "/api/v1/repos/ScottyLabs/study/push_mirrors"
    }

Plan: 7 to add, 1 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 7 to add, 1 to change, 0 to destroy.


  • ⏩ To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.05s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # forgejo_repository.cmu-study_study will be updated in-place ! resource "forgejo_repository" "cmu-study_study" { ! clone_url = "https://git.cmu.dev/ScottyLabs/study.git" -> (known after apply) ! created_at = "2026-09-18T22:52:34Z" -> (known after apply) ! description = "Labrador Boilerplate for Web" -> "A platform for creating and joining study groups" ! empty = false -> (known after apply) + external_tracker = (known after apply) + external_wiki = (known after apply) ! forks_count = 0 -> (known after apply) ! full_name = "ScottyLabs/study" -> (known after apply) ! html_url = "https://git.cmu.dev/ScottyLabs/study" -> (known after apply) id = 93 ! internal_tracker = { ! allow_only_contributors_to_track_time = true -> (known after apply) ! enable_issue_dependencies = true -> (known after apply) ! enable_time_tracker = true -> (known after apply) } -> (known after apply) + mirror_interval = (known after apply) ! mirror_updated = "0001-01-01T00:00:00Z" -> (known after apply) name = "study" ! open_issues_count = 0 -> (known after apply) ! open_pr_counter = 0 -> (known after apply) ! permissions = { ! admin = true -> (known after apply) ! pull = true -> (known after apply) ! push = true -> (known after apply) } -> (known after apply) ! release_counter = 0 -> (known after apply) ! size = 2234 -> (known after apply) ! ssh_url = "ssh://forgejo@git.cmu.dev/ScottyLabs/study.git" -> (known after apply) ! stars_count = 0 -> (known after apply) ! updated_at = "2026-09-18T22:52:36Z" -> (known after apply) ! watchers_count = 3 -> (known after apply) ! website = "https://study-group-finder-eight.vercel.app" -> "https://cmustudy.com" # (30 unchanged attributes hidden) } # forgejo_team_member.cmu-study_annadavi will be created + resource "forgejo_team_member" "cmu-study_annadavi" { + team_id = 43 + user = "annadavi" } # forgejo_team_repository.cmu-study_study will be created + resource "forgejo_team_repository" "cmu-study_study" { + owner = "ScottyLabs" + repository = "study" + team_id = 43 } # forgejo_team_repository.study_leads_study will be created + resource "forgejo_team_repository" "study_leads_study" { + owner = "ScottyLabs" + repository = "study" + team_id = 42 } # github_repository_deploy_key.cmu-study_study_mirror_deploy_key will be created + resource "github_repository_deploy_key" "cmu-study_study_mirror_deploy_key" { + etag = (known after apply) + id = (known after apply) + key = (known after apply) + read_only = false + repository = "study" + title = "Forgejo mirror" } # github_repository_ruleset.cmu-study_study_ruleset will be created + resource "github_repository_ruleset" "cmu-study_study_ruleset" { + enforcement = "active" + etag = (known after apply) + id = (known after apply) + name = "Default" + node_id = (known after apply) + repository = "study" + ruleset_id = (known after apply) + target = "branch" + bypass_actors { + actor_id = (known after apply) + actor_type = "DeployKey" + bypass_mode = "always" } + conditions { + ref_name { + exclude = [] + include = [ + "~ALL", ] } } + rules { + creation = true + deletion = true + non_fast_forward = true + update = true + update_allows_fetch_and_merge = false + pull_request { + allowed_merge_methods = (known after apply) + dismiss_stale_reviews_on_push = false + require_code_owner_review = false + require_last_push_approval = false + required_approving_review_count = 1 + required_review_thread_resolution = false } } } # restapi_object.cmu-study_study_kennel_webhook will be created + resource "restapi_object" "cmu-study_study_kennel_webhook" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = (sensitive value) + id = (known after apply) + id_attribute = "id" + path = "/api/v1/repos/ScottyLabs/study/hooks" + update_method = "PATCH" } # restapi_object.cmu-study_study_push_mirror will be created + resource "restapi_object" "cmu-study_study_push_mirror" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = jsonencode( { + interval = "8h0m0s" + remote_address = "git@github.com:ScottyLabs/study.git" + sync_on_commit = true + use_ssh = true } ) + id = (known after apply) + id_attribute = "remote_name" + path = "/api/v1/repos/ScottyLabs/study/push_mirrors" } Plan: 7 to add, 1 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 7 to add, 1 to change, 0 to destroy. --- * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis apply -p git

atlantis apply -p git

Ran Apply for project: git dir: tofu/git workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.05s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.cmu-study_annadavi: Creating...
forgejo_team_member.cmu-study_annadavi: Creation complete after 0s
forgejo_repository.cmu-study_study: Modifying... [name=study]
forgejo_repository.cmu-study_study: Modifications complete after 0s [name=study]
forgejo_team_repository.cmu-study_study: Creating...
forgejo_team_repository.cmu-study_study: Creation complete after 0s
forgejo_team_repository.study_leads_study: Creating...
forgejo_team_repository.study_leads_study: Creation complete after 0s
restapi_object.cmu-study_study_push_mirror: Creating...
restapi_object.cmu-study_study_kennel_webhook: Creating...
restapi_object.cmu-study_study_kennel_webhook: Creation complete after 0s [id=33]
restapi_object.cmu-study_study_push_mirror: Creation complete after 0s [id=remote_mirror_Lo5Rz3Yz7Zp]
github_repository_deploy_key.cmu-study_study_mirror_deploy_key: Creating...
github_repository_deploy_key.cmu-study_study_mirror_deploy_key: Creation complete after 2s [id=study:163760029]
github_repository_ruleset.cmu-study_study_ruleset: Creating...
github_repository_ruleset.cmu-study_study_ruleset: Creation complete after 0s [id=23686597]

Apply complete! Resources: 7 added, 1 changed, 0 destroyed.
Ran Apply for project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.05s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.cmu-study_annadavi: Creating... forgejo_team_member.cmu-study_annadavi: Creation complete after 0s forgejo_repository.cmu-study_study: Modifying... [name=study] forgejo_repository.cmu-study_study: Modifications complete after 0s [name=study] forgejo_team_repository.cmu-study_study: Creating... forgejo_team_repository.cmu-study_study: Creation complete after 0s forgejo_team_repository.study_leads_study: Creating... forgejo_team_repository.study_leads_study: Creation complete after 0s restapi_object.cmu-study_study_push_mirror: Creating... restapi_object.cmu-study_study_kennel_webhook: Creating... restapi_object.cmu-study_study_kennel_webhook: Creation complete after 0s [id=33] restapi_object.cmu-study_study_push_mirror: Creation complete after 0s [id=remote_mirror_Lo5Rz3Yz7Zp] github_repository_deploy_key.cmu-study_study_mirror_deploy_key: Creating... github_repository_deploy_key.cmu-study_study_mirror_deploy_key: Creation complete after 2s [id=study:163760029] github_repository_ruleset.cmu-study_study_ruleset: Creating... github_repository_ruleset.cmu-study_study_ruleset: Creation complete after 0s [id=23686597] Apply complete! Resources: 7 added, 1 changed, 0 destroyed. ``` </details>
anish merged commit ac8ebfb364 into main 2026-09-19 00:21:01 +00:00
anish referenced this pull request from a commit 2026-09-19 00:21:03 +00:00

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/matrix_bridges workspace: default
  • dir: tofu/openbao workspace: default
  • dir: tofu/posthog workspace: default
  • dir: tofu/sentry workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/matrix_bridges` workspace: `default` - dir: `tofu/openbao` workspace: `default` - dir: `tofu/posthog` workspace: `default` - dir: `tofu/sentry` workspace: `default` - dir: `tofu/slack` workspace: `default`
brightz referenced this pull request from a commit 2026-09-26 22:03:28 +00:00
Sign in to join this conversation.
No description provided.