chore: add akhanal2 to slai #210
No reviewers
Labels
No labels
needs/docs
needs/research
needs/upstream
type/bug
type/external
type/feature
type/refactor
urgency
high
urgency
immediate
urgency
low
urgency
medium
No milestone
No project
No assignees
2 participants
Notifications
Due date
No due date set.
Dependencies
No dependencies set
Reference
ScottyLabs/governance!210
Loading…
Reference in a new issue
No description provided.
Delete branch "akhanal2/governance:profile_chore/add-akhanal2-to-slai"
Deleting a branch is permanent. Although the deleted branch may continue to exist for a short time before it actually gets removed, it CANNOT be undone in most cases. Continue?
Adds akhanal2 to the SLAI team membership list.
Ran Plan for 6 projects:
google_groupsdir:tofu/google_groupsworkspace:defaultgitdir:tofu/gitworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgoogle_playdir:tofu/google_playworkspace:defaultslackdir:tofu/slackworkspace:defaultkeycloakdir:tofu/keycloakworkspace:default1. project:
google_groupsdir:tofu/google_groupsworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.2. project:
gitdir:tofu/gitworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.3. project:
discorddir:tofu/discordworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.4. project:
google_playdir:tofu/google_playworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.5. project:
slackdir:tofu/slackworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.6. project:
keycloakdir:tofu/keycloakworkspace:defaultPlan Failed: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request.
Once the lock is released, comment
atlantis planhere to re-plan.Plan Summary
6 projects, 0 with changes, 0 with no changes, 6 failed
View command line instructions
Checkout
From your project repository, check out a new branch and test the changes.Merge
Merge the changes and update on Forgejo.Warning: The "Autodetect manual merge" setting is not enabled for this repository, you will have to mark this pull request as manually merged afterwards.