feat(slai): add adamzhu to members #204
No reviewers
Labels
No labels
needs/docs
needs/research
needs/upstream
type/bug
type/external
type/feature
type/refactor
urgency
high
urgency
immediate
urgency
low
urgency
medium
No milestone
No project
No assignees
3 participants
Notifications
Due date
No due date set.
Dependencies
No dependencies set
Reference
ScottyLabs/governance!204
Loading…
Reference in a new issue
No description provided.
Delete branch "adamzhu/governance:add-adamzhu-slai"
Deleting a branch is permanent. Although the deleted branch may continue to exist for a short time before it actually gets removed, it CANNOT be undone in most cases. Continue?
Adding myself (adamzhu) to the ScottyLabs AI team members list per Krish's onboarding email.
Ran Plan for 6 projects:
slackdir:tofu/slackworkspace:defaultgoogle_playdir:tofu/google_playworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 0 to destroy.
2. project:
google_playdir:tofu/google_playworkspace:default3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 0 to destroy.
4. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 0 to add, 5 to change, 0 to destroy.
5. project:
discorddir:tofu/discordworkspace:defaultShow Output
Plan: 1 to add, 0 to change, 0 to destroy.
6. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 2 to add, 0 to change, 0 to destroy.
Plan Summary
6 projects, 5 with changes, 1 with no changes, 0 failed
atlantis unlock
All Atlantis locks for this PR have been unlocked and plans discarded
View command line instructions
Checkout
From your project repository, check out a new branch and test the changes.Merge
Merge the changes and update on Forgejo.Warning: The "Autodetect manual merge" setting is not enabled for this repository, you will have to mark this pull request as manually merged afterwards.