chore: add cmu-courses member #199

Merged
anish merged 2 commits from gabrielx/governance:main into main 2026-09-18 06:18:33 +00:00
Member
No description provided.
chore: add cmu-courses-member
Some checks failed
atlantis/plan Plan failed.
kennel/build skipped: no services or static sites defined
CI / check (pull_request) Blocked by required conditions
CI / check-1 (pull_request) Blocked by required conditions
CI / build (pull_request) Blocked by required conditions
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
Validate / Governance (pull_request_target) Successful in 4m19s
d705e2816e
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
All checks were successful
atlantis/plan: slack Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_play Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 3 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
atlantis/plan: git Plan: 2 to add, 2 to change, 0 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
5252d21320

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.12s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-courses_ch0_gabrielx will be created
+ resource "null_resource" "slack_cmu-courses_ch0_gabrielx" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C0150RGAG1L"
          + "user"    = "U0C129SHN5R"
        }
    }

  # null_resource.slack_hub_gabrielx will be created
+ resource "null_resource" "slack_hub_gabrielx" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0C129SHN5R"
        }
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 2 to add, 0 to change, 0 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
   Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/199/default/crates/governance-schema)
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/199/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/199/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/199/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 9.12s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # googleplay_app_iam.courses_gabrielx will be created
+ resource "googleplay_app_iam" "courses_gabrielx" {
+       app_id               = "courses.cmu.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "gabrielx@andrew.cmu.edu"
    }

  # googleplay_user.gabrielx will be created
+ resource "googleplay_user" "gabrielx" {
+       email                = "gabrielx@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 2 to add, 0 to change, 0 to destroy.


3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.16s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_gabrielx will be created
+ resource "google_cloud_identity_group_membership" "tech_gabrielx" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "gabrielx@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.13s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-courses will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-courses" {
        id       = "scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77"
!       members  = [
+           "gabrielx",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-courses_courses will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-courses_courses" {
        id       = "scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406"
!       members  = [
+           "gabrielx",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "gabrielx",
            # (31 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 3 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 3 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.15s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.gabrielx will be created
+ resource "discord_member_roles" "gabrielx" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "1402407381542899852"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1500016758700904488"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 9.11s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # forgejo_repository.cmu-housing_housing will be updated in-place
! resource "forgejo_repository" "cmu-housing_housing" {
!       clone_url                   = "https://git.cmu.dev/ScottyLabs/housing.git" -> (known after apply)
!       created_at                  = "2026-08-08T10:24:09Z" -> (known after apply)
!       empty                       = false -> (known after apply)
+       external_tracker            = (known after apply)
!       external_wiki               = {
!           external_wiki_url = "https://docs.scottylabs.org/housing/contributing/" -> (known after apply)
        } -> (known after apply)
!       forks_count                 = 2 -> (known after apply)
!       full_name                   = "ScottyLabs/housing" -> (known after apply)
!       html_url                    = "https://git.cmu.dev/ScottyLabs/housing" -> (known after apply)
        id                          = 23
!       internal_tracker            = {
!           allow_only_contributors_to_track_time = true -> (known after apply)
!           enable_issue_dependencies             = true -> (known after apply)
!           enable_time_tracker                   = true -> (known after apply)
        } -> (known after apply)
+       mirror_interval             = (known after apply)
!       mirror_updated              = "0001-01-01T00:00:00Z" -> (known after apply)
        name                        = "housing"
!       open_issues_count           = 13 -> (known after apply)
!       open_pr_counter             = 2 -> (known after apply)
!       permissions                 = {
!           admin = true -> (known after apply)
!           pull  = true -> (known after apply)
!           push  = true -> (known after apply)
        } -> (known after apply)
!       release_counter             = 0 -> (known after apply)
!       size                        = 100032 -> (known after apply)
!       ssh_url                     = "ssh://forgejo@git.cmu.dev/ScottyLabs/housing.git" -> (known after apply)
!       stars_count                 = 0 -> (known after apply)
!       updated_at                  = "2026-09-18T04:34:42Z" -> (known after apply)
!       watchers_count              = 13 -> (known after apply)
        # (32 unchanged attributes hidden)
    }

  # forgejo_team_member.cmu-courses_gabrielx will be created
+ resource "forgejo_team_member" "cmu-courses_gabrielx" {
+       team_id = 12
+       user    = "gabrielx"
    }

  # github_repository.slai_cmugpt_agent will be updated in-place
! resource "github_repository" "slai_cmugpt_agent" {
        id                                      = "cmugpt-agent"
        name                                    = "cmugpt-agent"
        # (39 unchanged attributes hidden)

-       template {
-           include_all_branches = false -> null
-           owner                = "ScottyLabs" -> null
-           repository           = "python-template" -> null
        }

        # (1 unchanged block hidden)
    }

  # github_team_membership.cmu-courses_gabrielx will be created
+ resource "github_team_membership" "cmu-courses_gabrielx" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "3643907"
+       username = "gabriel-xu"
    }

Plan: 2 to add, 2 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 2 to change, 0 to destroy.


Plan Summary

6 projects, 6 with changes, 0 with no changes, 0 failed

  • To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.12s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_cmu-courses_ch0_gabrielx will be created + resource "null_resource" "slack_cmu-courses_ch0_gabrielx" { + id = (known after apply) + triggers = { + "channel" = "C0150RGAG1L" + "user" = "U0C129SHN5R" } } # null_resource.slack_hub_gabrielx will be created + resource "null_resource" "slack_hub_gabrielx" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0C129SHN5R" } } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/199/default/crates/governance-schema) Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/199/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/199/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/199/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 9.12s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # googleplay_app_iam.courses_gabrielx will be created + resource "googleplay_app_iam" "courses_gabrielx" { + app_id = "courses.cmu.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "gabrielx@andrew.cmu.edu" } # googleplay_user.gabrielx will be created + resource "googleplay_user" "gabrielx" { + email = "gabrielx@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.16s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_gabrielx will be created + resource "google_cloud_identity_group_membership" "tech_gabrielx" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "gabrielx@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.13s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-courses will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-courses" { id = "scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77" ! members = [ + "gabrielx", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-courses_courses will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-courses_courses" { id = "scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406" ! members = [ + "gabrielx", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "gabrielx", # (31 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 3 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 3 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.15s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.gabrielx will be created + resource "discord_member_roles" "gabrielx" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "1402407381542899852" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1500016758700904488" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 9.11s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # forgejo_repository.cmu-housing_housing will be updated in-place ! resource "forgejo_repository" "cmu-housing_housing" { ! clone_url = "https://git.cmu.dev/ScottyLabs/housing.git" -> (known after apply) ! created_at = "2026-08-08T10:24:09Z" -> (known after apply) ! empty = false -> (known after apply) + external_tracker = (known after apply) ! external_wiki = { ! external_wiki_url = "https://docs.scottylabs.org/housing/contributing/" -> (known after apply) } -> (known after apply) ! forks_count = 2 -> (known after apply) ! full_name = "ScottyLabs/housing" -> (known after apply) ! html_url = "https://git.cmu.dev/ScottyLabs/housing" -> (known after apply) id = 23 ! internal_tracker = { ! allow_only_contributors_to_track_time = true -> (known after apply) ! enable_issue_dependencies = true -> (known after apply) ! enable_time_tracker = true -> (known after apply) } -> (known after apply) + mirror_interval = (known after apply) ! mirror_updated = "0001-01-01T00:00:00Z" -> (known after apply) name = "housing" ! open_issues_count = 13 -> (known after apply) ! open_pr_counter = 2 -> (known after apply) ! permissions = { ! admin = true -> (known after apply) ! pull = true -> (known after apply) ! push = true -> (known after apply) } -> (known after apply) ! release_counter = 0 -> (known after apply) ! size = 100032 -> (known after apply) ! ssh_url = "ssh://forgejo@git.cmu.dev/ScottyLabs/housing.git" -> (known after apply) ! stars_count = 0 -> (known after apply) ! updated_at = "2026-09-18T04:34:42Z" -> (known after apply) ! watchers_count = 13 -> (known after apply) # (32 unchanged attributes hidden) } # forgejo_team_member.cmu-courses_gabrielx will be created + resource "forgejo_team_member" "cmu-courses_gabrielx" { + team_id = 12 + user = "gabrielx" } # github_repository.slai_cmugpt_agent will be updated in-place ! resource "github_repository" "slai_cmugpt_agent" { id = "cmugpt-agent" name = "cmugpt-agent" # (39 unchanged attributes hidden) - template { - include_all_branches = false -> null - owner = "ScottyLabs" -> null - repository = "python-template" -> null } # (1 unchanged block hidden) } # github_team_membership.cmu-courses_gabrielx will be created + resource "github_team_membership" "cmu-courses_gabrielx" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "3643907" + username = "gabriel-xu" } Plan: 2 to add, 2 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 2 to change, 0 to destroy. --- ### Plan Summary 6 projects, 6 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis apply

atlantis apply

Ran Apply for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: google_play dir: tofu/google_play workspace: default
  5. project: keycloak dir: tofu/keycloak workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.26s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_hub_gabrielx: Creating...
null_resource.slack_cmu-courses_ch0_gabrielx: Creating...
null_resource.slack_cmu-courses_ch0_gabrielx: Provisioning with 'local-exec'...
null_resource.slack_cmu-courses_ch0_gabrielx (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0150RGAG1L --user U0C129SHN5R"]
null_resource.slack_hub_gabrielx: Provisioning with 'local-exec'...
null_resource.slack_hub_gabrielx (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C129SHN5R"]
null_resource.slack_cmu-courses_ch0_gabrielx: Creation complete after 0s [id=2637814645010454484]
null_resource.slack_hub_gabrielx: Creation complete after 0s [id=6954671545047364331]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

2. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.26s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.gabrielx: Creating...
discord_member_roles.gabrielx: Creation complete after 0s [id=1353118209024331948:1402407381542899852]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.22s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_gabrielx: Creating...
google_cloud_identity_group_membership.tech_gabrielx: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/118445324624041071464]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

4. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on shared package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.30s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
googleplay_user.gabrielx: Creating...
googleplay_user.gabrielx: Creation complete after 1s [id=gabrielx@andrew.cmu.edu]
googleplay_app_iam.courses_gabrielx: Creating...
googleplay_app_iam.courses_gabrielx: Creation complete after 1s

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

5. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.27s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cmu-courses: Modifying... [id=scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77]
keycloak_group_memberships.project_cmu-courses_courses: Modifying... [id=scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_cmu-courses: Modifications complete after 2s [id=scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77]
keycloak_group_memberships.project_cmu-courses_courses: Modifications complete after 2s [id=scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406]
keycloak_group_memberships.projects: Modifications complete after 2s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 3 changed, 0 destroyed.

6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.23s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.cmu-courses_gabrielx: Creating...
forgejo_team_member.cmu-courses_gabrielx: Creation complete after 0s
forgejo_repository.cmu-housing_housing: Modifying... [name=housing]
forgejo_repository.cmu-housing_housing: Modifications complete after 0s [name=housing]
github_team_membership.cmu-courses_gabrielx: Creating...
github_repository.slai_cmugpt_agent: Modifying... [id=cmugpt-agent]
github_team_membership.cmu-courses_gabrielx: Creation complete after 3s [id=3643907:gabriel-xu]
github_repository.slai_cmugpt_agent: Modifications complete after 4s [id=cmugpt-agent]

Apply complete! Resources: 2 added, 2 changed, 0 destroyed.

Apply Summary

6 projects, 6 successful, 0 failed, 0 errored

Ran Apply for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.26s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_hub_gabrielx: Creating... null_resource.slack_cmu-courses_ch0_gabrielx: Creating... null_resource.slack_cmu-courses_ch0_gabrielx: Provisioning with 'local-exec'... null_resource.slack_cmu-courses_ch0_gabrielx (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C0150RGAG1L --user U0C129SHN5R"] null_resource.slack_hub_gabrielx: Provisioning with 'local-exec'... null_resource.slack_hub_gabrielx (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0C129SHN5R"] null_resource.slack_cmu-courses_ch0_gabrielx: Creation complete after 0s [id=2637814645010454484] null_resource.slack_hub_gabrielx: Creation complete after 0s [id=6954671545047364331] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.26s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.gabrielx: Creating... discord_member_roles.gabrielx: Creation complete after 0s [id=1353118209024331948:1402407381542899852] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.22s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_gabrielx: Creating... google_cloud_identity_group_membership.tech_gabrielx: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/118445324624041071464] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on shared package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.30s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ googleplay_user.gabrielx: Creating... googleplay_user.gabrielx: Creation complete after 1s [id=gabrielx@andrew.cmu.edu] googleplay_app_iam.courses_gabrielx: Creating... googleplay_app_iam.courses_gabrielx: Creation complete after 1s Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.27s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cmu-courses: Modifying... [id=scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77] keycloak_group_memberships.project_cmu-courses_courses: Modifying... [id=scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_cmu-courses: Modifications complete after 2s [id=scottylabs/group-memberships/bbf3e8ad-e627-41cb-80e9-2c48768a6a77] keycloak_group_memberships.project_cmu-courses_courses: Modifications complete after 2s [id=scottylabs/group-memberships/3f1ea03c-766b-4265-9a4a-b63463a5e406] keycloak_group_memberships.projects: Modifications complete after 2s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 3 changed, 0 destroyed. ``` </details> --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.23s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.cmu-courses_gabrielx: Creating... forgejo_team_member.cmu-courses_gabrielx: Creation complete after 0s forgejo_repository.cmu-housing_housing: Modifying... [name=housing] forgejo_repository.cmu-housing_housing: Modifications complete after 0s [name=housing] github_team_membership.cmu-courses_gabrielx: Creating... github_repository.slai_cmugpt_agent: Modifying... [id=cmugpt-agent] github_team_membership.cmu-courses_gabrielx: Creation complete after 3s [id=3643907:gabriel-xu] github_repository.slai_cmugpt_agent: Modifications complete after 4s [id=cmugpt-agent] Apply complete! Resources: 2 added, 2 changed, 0 destroyed. ``` </details> --- ### Apply Summary 6 projects, 6 successful, 0 failed, 0 errored
anish merged commit 16b67c3972 into main 2026-09-18 06:18:33 +00:00

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
anish referenced this pull request from a commit 2026-09-18 06:18:34 +00:00
All checks were successful
atlantis/plan: slack Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_play Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 3 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
atlantis/plan: git Plan: 2 to add, 2 to change, 0 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
Some workflows are waiting to be reviewed.
Sign in to join this conversation.
No description provided.