feat: add labrador and outreach committee channels #194

Merged
anish merged 2 commits from committees into main 2026-09-15 05:11:51 +00:00
Owner
No description provided.
feat: add labrador and outreach committee channels
Some checks failed
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 4m50s
CI / check-1 (pull_request) Failing after 6m4s
CI / check (pull_request) Failing after 0s
7a5a87bf57
Signed-off-by: Anish Pallati <i@anish.land>
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
All checks were successful
kennel/build skipped: no services or static sites defined
atlantis/plan: slack Plan: 8 to add, 0 to change, 0 to destroy.
atlantis/plan: matrix_bridges Plan: 4 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 28 to change, 0 to destroy.
atlantis/plan: discord Plan: 0 to add, 4 to change, 0 to destroy.
atlantis/plan: git Plan: 44 to add, 2 to change, 0 to destroy.
atlantis/plan 5/5 projects planned successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply: matrix_bridges Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply 5/5 projects applied successfully.
7cf53fbc5a

Ran Plan for 5 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: matrix_bridges dir: tofu/matrix_bridges workspace: default
  3. project: keycloak dir: tofu/keycloak workspace: default
  4. project: discord dir: tofu/discord workspace: default
  5. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
   Compiling proc-macro2 v1.0.106
   Compiling quote v1.0.45
   Compiling unicode-ident v1.0.24
   Compiling serde_core v1.0.228
   Compiling shlex v1.3.0
   Compiling find-msvc-tools v0.1.9
   Compiling libc v0.2.183
   Compiling cc v1.2.57
   Compiling cfg-if v1.0.4
   Compiling zmij v1.0.21
   Compiling syn v2.0.117
   Compiling itoa v1.0.18
   Compiling ring v0.17.14
   Compiling zeroize v1.8.2
   Compiling serde_json v1.0.149
   Compiling serde v1.0.228
   Compiling rustls-pki-types v1.14.0
   Compiling getrandom v0.2.17
   Compiling ref-cast v1.0.25
   Compiling untrusted v0.9.0
   Compiling crc32fast v1.5.0
   Compiling httparse v1.10.1
   Compiling memchr v2.8.0
   Compiling serde_derive_internals v0.29.1
   Compiling serde_derive v1.0.228
   Compiling ref-cast-impl v1.0.25
   Compiling simd-adler32 v0.3.8
   Compiling bytes v1.11.1
   Compiling rustls v0.23.37
   Compiling adler2 v2.0.1
   Compiling log v0.4.29
   Compiling miniz_oxide v0.8.9
   Compiling http v1.4.0
   Compiling schemars_derive v1.2.1
   Compiling rustls-webpki v0.103.13
   Compiling thiserror v2.0.18
   Compiling dyn-clone v1.0.20
   Compiling base64 v0.22.1
   Compiling once_cell v1.21.4
   Compiling winnow v1.0.0
   Compiling subtle v2.6.1
   Compiling utf8parse v0.2.2
   Compiling anstyle-parse v1.0.0
   Compiling toml_parser v1.1.0+spec-1.1.0
   Compiling ureq-proto v0.6.0
   Compiling schemars v1.2.1
   Compiling flate2 v1.1.9
   Compiling thiserror-impl v2.0.18
   Compiling webpki-roots v1.0.6
   Compiling serde_spanned v1.1.0
   Compiling toml_datetime v1.1.0+spec-1.1.0
   Compiling is_terminal_polyfill v1.70.2
   Compiling utf8-zero v0.8.1
   Compiling anstyle v1.0.14
   Compiling toml_writer v1.1.0+spec-1.1.0
   Compiling percent-encoding v2.3.2
   Compiling anstyle-query v1.1.5
   Compiling colorchoice v1.0.5
   Compiling anstream v1.0.0
   Compiling toml v1.1.0+spec-1.1.0
   Compiling ureq v3.3.0
   Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/194/default/crates/governance-schema)
   Compiling clap_lex v1.1.0
   Compiling anyhow v1.0.104
   Compiling heck v0.5.0
   Compiling strsim v0.11.1
   Compiling clap_builder v4.6.0
   Compiling clap_derive v4.6.0
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/194/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/194/default/crates/governance-tfgen)
   Compiling clap v4.6.0
   Compiling dotenvy v0.15.7
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/194/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 45.60s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_cbp_ch0_anish will be created
+ resource "null_resource" "slack_cbp_ch0_anish" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08S56CJ8R2"
          + "user"    = "U07FPJKFB5E"
        }
    }

  # null_resource.slack_cbp_ch0_nikhilr will be created
+ resource "null_resource" "slack_cbp_ch0_nikhilr" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08S56CJ8R2"
          + "user"    = "U0AMBE2QKBM"
        }
    }

  # null_resource.slack_cbp_ch0_xboxbedrock will be created
+ resource "null_resource" "slack_cbp_ch0_xboxbedrock" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08S56CJ8R2"
          + "user"    = "U0AQRKYLGHK"
        }
    }

  # null_resource.slack_cmu-cal_ch0_anish will be created
+ resource "null_resource" "slack_cmu-cal_ch0_anish" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C05TNBPHVTL"
          + "user"    = "U07FPJKFB5E"
        }
    }

  # null_resource.slack_lost-and-found_ch0_anish will be created
+ resource "null_resource" "slack_lost-and-found_ch0_anish" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C01BAADS8U8"
          + "user"    = "U07FPJKFB5E"
        }
    }

  # null_resource.slack_slai_ch0_anish will be created
+ resource "null_resource" "slack_slai_ch0_anish" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08SRUGUQ05"
          + "user"    = "U07FPJKFB5E"
        }
    }

  # null_resource.slack_slai_ch0_xboxbedrock will be created
+ resource "null_resource" "slack_slai_ch0_xboxbedrock" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08SRUGUQ05"
          + "user"    = "U0AQRKYLGHK"
        }
    }

  # null_resource.slack_uia_ch0_anish will be created
+ resource "null_resource" "slack_uia_ch0_anish" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C08MAT652CE"
          + "user"    = "U07FPJKFB5E"
        }
    }

Plan: 8 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 8 to add, 0 to change, 0 to destroy.


2. project: matrix_bridges dir: tofu/matrix_bridges workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 45.66s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_bridge_login_org_labrador will be created
+ resource "null_resource" "slack_bridge_login_org_labrador" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C05TG63EM9B"
        }
    }

  # null_resource.slack_bridge_login_org_outreach will be created
+ resource "null_resource" "slack_bridge_login_org_outreach" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C041U79AZAT"
        }
    }

  # synapse_mautrix_slack_link.org_labrador will be created
+ resource "synapse_mautrix_slack_link" "org_labrador" {
+       discord_channel_id       = "1549280062581252187"
+       discord_relay_configured = (known after apply)
+       id                       = (known after apply)
+       project_name             = "Labrador"
+       project_slug             = "labrador"
+       room_alias               = (known after apply)
+       room_id                  = (known after apply)
+       slack_bridged            = (known after apply)
+       slack_channel_id         = "C05TG63EM9B"
+       team_name                = "ScottyLabs"
+       team_slug                = "org"
    }

  # synapse_mautrix_slack_link.org_outreach will be created
+ resource "synapse_mautrix_slack_link" "org_outreach" {
+       discord_channel_id       = "1549280019686101182"
+       discord_relay_configured = (known after apply)
+       id                       = (known after apply)
+       project_name             = "Outreach"
+       project_slug             = "outreach"
+       room_alias               = (known after apply)
+       room_id                  = (known after apply)
+       slack_bridged            = (known after apply)
+       slack_channel_id         = "C041U79AZAT"
+       team_name                = "ScottyLabs"
+       team_slug                = "org"
    }

Plan: 4 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p matrix_bridges
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p matrix_bridges
    

Plan: 4 to add, 0 to change, 0 to destroy.


3. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 45.62s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cbp_bus-sign will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_bus-sign" {
        id       = "scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b"
!       members  = [
+           "apallati",
+           "nramanuj",
+           "spola2",
            # (3 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_bus-sign_admins will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_bus-sign_admins" {
        id       = "scottylabs/group-memberships/100e7090-c4a4-47f2-b0ad-8eb81dd8c8c8"
!       members  = [
+           "apallati",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_dalmatian will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_dalmatian" {
        id       = "scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153"
!       members  = [
+           "apallati",
+           "nramanuj",
+           "spola2",
            # (3 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_dalmatian_admins will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_dalmatian_admins" {
        id       = "scottylabs/group-memberships/5467214e-843c-488a-85fb-2ca5fa77d7d7"
!       members  = [
+           "apallati",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_discord-verify will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_discord-verify" {
        id       = "scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745"
!       members  = [
+           "apallati",
+           "nramanuj",
+           "spola2",
            # (3 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_discord-verify_admins will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_discord-verify_admins" {
        id       = "scottylabs/group-memberships/a2abc505-3248-42f8-9513-d3db72be5d40"
!       members  = [
+           "apallati",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_groupme-mirror will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_groupme-mirror" {
        id       = "scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39"
!       members  = [
+           "apallati",
+           "nramanuj",
+           "spola2",
            # (3 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cbp_groupme-mirror_admins will be updated in-place
! resource "keycloak_group_memberships" "project_cbp_groupme-mirror_admins" {
        id       = "scottylabs/group-memberships/c4c2efc2-d322-4193-8279-699e397522b0"
!       members  = [
+           "apallati",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-cal_cal will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-cal_cal" {
        id       = "scottylabs/group-memberships/3800a836-0a57-40b7-839f-650a335bfccc"
!       members  = [
+           "apallati",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-cal_cal_admins will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-cal_cal_admins" {
        id       = "scottylabs/group-memberships/5ca15cf1-036f-4644-b45e-8ca25ce9b2af"
!       members  = [
+           "apallati",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-housing_housing" {
        id       = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d"
!       members  = [
+           "apallati",
            # (10 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_cmu-housing_housing_admins will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-housing_housing_admins" {
        id       = "scottylabs/group-memberships/7c932acc-ac1e-420b-8040-9c3af456b597"
!       members  = [
+           "apallati",
            # (4 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_lost-and-found_lost-and-found-v2 will be updated in-place
! resource "keycloak_group_memberships" "project_lost-and-found_lost-and-found-v2" {
        id       = "scottylabs/group-memberships/c14b4061-e3db-42c9-9b9e-a0c0a9747ce4"
!       members  = [
+           "apallati",
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_lost-and-found_lost-and-found-v2_admins will be updated in-place
! resource "keycloak_group_memberships" "project_lost-and-found_lost-and-found-v2_admins" {
        id       = "scottylabs/group-memberships/7879878f-16a8-4b07-835e-81a8a9e7b02f"
!       members  = [
+           "apallati",
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-agent will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-agent" {
        id       = "scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423"
!       members  = [
+           "apallati",
+           "spola2",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-agent_admins will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-agent_admins" {
        id       = "scottylabs/group-memberships/f4e82f9b-769d-4f38-a7f2-e214189d9269"
!       members  = [
+           "apallati",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-mcp-server will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server" {
        id       = "scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8"
!       members  = [
+           "apallati",
+           "spola2",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-mcp-server_admins will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server_admins" {
        id       = "scottylabs/group-memberships/8f84f9ec-e78a-41f6-9ce6-097fce273298"
!       members  = [
+           "apallati",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-sms-surface will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface" {
        id       = "scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43"
!       members  = [
+           "apallati",
+           "spola2",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-sms-surface_admins will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface_admins" {
        id       = "scottylabs/group-memberships/34404457-985f-4a61-b2f8-4754c2eb3293"
!       members  = [
+           "apallati",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-surface will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-surface" {
        id       = "scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775"
!       members  = [
+           "apallati",
+           "spola2",
            # (5 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_slai_cmugpt-surface_admins will be updated in-place
! resource "keycloak_group_memberships" "project_slai_cmugpt-surface_admins" {
        id       = "scottylabs/group-memberships/4f3531db-8075-480e-bbf4-253e98783b56"
!       members  = [
+           "apallati",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_tartan-vote will be updated in-place
! resource "keycloak_group_memberships" "project_tartan-vote" {
        id       = "scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c"
!       members  = [
+           "apallati",
+           "maxwen",
+           "spola2",
            # (2 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_tartan-vote_admins will be updated in-place
! resource "keycloak_group_memberships" "project_tartan-vote_admins" {
        id       = "scottylabs/group-memberships/cee98bbd-fb26-4527-ac5d-8a8587de2416"
!       members  = [
+           "apallati",
            # (1 unchanged element hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_uia_components will be updated in-place
! resource "keycloak_group_memberships" "project_uia_components" {
        id       = "scottylabs/group-memberships/95465a93-b14a-459d-85b5-12f8b0fe7fd2"
!       members  = [
+           "apallati",
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_uia_components_admins will be updated in-place
! resource "keycloak_group_memberships" "project_uia_components_admins" {
        id       = "scottylabs/group-memberships/75073d65-02c5-405b-bc54-b4036dbda934"
!       members  = [
+           "apallati",
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_uia_scottylabs-org will be updated in-place
! resource "keycloak_group_memberships" "project_uia_scottylabs-org" {
        id       = "scottylabs/group-memberships/f7ea42f9-6007-4a5d-b46b-e5ac1d3e2bb8"
!       members  = [
+           "apallati",
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.project_uia_scottylabs-org_admins will be updated in-place
! resource "keycloak_group_memberships" "project_uia_scottylabs-org_admins" {
        id       = "scottylabs/group-memberships/5786d2ba-5816-4ce2-8f04-d76958c4ee91"
!       members  = [
+           "apallati",
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 28 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 28 to change, 0 to destroy.


4. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 45.61s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # discord_member_roles.anish will be updated in-place
! resource "discord_member_roles" "anish" {
        id        = "1353118209024331948:315249559473225729"
        # (2 unchanged attributes hidden)

+       role {
+           has_role = true
+           role_id  = "1494773174334783499"
        }
+       role {
+           has_role = true
+           role_id  = "1504199977000505445"
        }
+       role {
+           has_role = true
+           role_id  = "1505078357321191535"
        }
+       role {
+           has_role = true
+           role_id  = "1508601984255524885"
        }
+       role {
+           has_role = true
+           role_id  = "1513653499957874871"
        }
+       role {
+           has_role = true
+           role_id  = "1526500755446698124"
        }
+       role {
+           has_role = true
+           role_id  = "1526511044762271914"
        }

        # (9 unchanged blocks hidden)
    }

  # discord_member_roles.nikhilr will be updated in-place
! resource "discord_member_roles" "nikhilr" {
        id        = "1353118209024331948:795153900553764874"
        # (2 unchanged attributes hidden)

+       role {
+           has_role = true
+           role_id  = "1494773174334783499"
        }

        # (3 unchanged blocks hidden)
    }

  # discord_member_roles.tentype will be updated in-place
! resource "discord_member_roles" "tentype" {
        id        = "1353118209024331948:637528457508552705"
        # (2 unchanged attributes hidden)

+       role {
+           has_role = true
+           role_id  = "1508601984255524885"
        }

        # (4 unchanged blocks hidden)
    }

  # discord_member_roles.xboxbedrock will be updated in-place
! resource "discord_member_roles" "xboxbedrock" {
        id        = "1353118209024331948:712398210345205871"
        # (2 unchanged attributes hidden)

+       role {
+           has_role = true
+           role_id  = "1494773174334783499"
        }
+       role {
+           has_role = true
+           role_id  = "1505078357321191535"
        }
+       role {
+           has_role = true
+           role_id  = "1508601984255524885"
        }

        # (6 unchanged blocks hidden)
    }

Plan: 0 to add, 4 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 0 to add, 4 to change, 0 to destroy.


5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 45.65s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # forgejo_team.components_leads will be created
+ resource "forgejo_team" "components_leads" {
+       can_create_org_repo       = false
+       description               = "Components leads"
+       id                        = (known after apply)
+       includes_all_repositories = false
+       name                      = "components-leads"
+       organization              = (known after apply)
+       organization_id           = 3
+       permission                = "admin"
+       units_map                 = {
          + "repo.actions"    = "admin"
          + "repo.code"       = "admin"
          + "repo.ext_issues" = "admin"
          + "repo.ext_wiki"   = "admin"
          + "repo.issues"     = "admin"
          + "repo.packages"   = "admin"
          + "repo.projects"   = "admin"
          + "repo.pulls"      = "admin"
          + "repo.releases"   = "admin"
          + "repo.wiki"       = "admin"
        }
    }

  # forgejo_team.lost-and-found-v2_leads will be created
+ resource "forgejo_team" "lost-and-found-v2_leads" {
+       can_create_org_repo       = false
+       description               = "Lost and Found leads"
+       id                        = (known after apply)
+       includes_all_repositories = false
+       name                      = "lost-and-found-v2-leads"
+       organization              = (known after apply)
+       organization_id           = 3
+       permission                = "admin"
+       units_map                 = {
          + "repo.actions"    = "admin"
          + "repo.code"       = "admin"
          + "repo.ext_issues" = "admin"
          + "repo.ext_wiki"   = "admin"
          + "repo.issues"     = "admin"
          + "repo.packages"   = "admin"
          + "repo.projects"   = "admin"
          + "repo.pulls"      = "admin"
          + "repo.releases"   = "admin"
          + "repo.wiki"       = "admin"
        }
    }

  # forgejo_team.scottylabs-org_leads will be created
+ resource "forgejo_team" "scottylabs-org_leads" {
+       can_create_org_repo       = false
+       description               = "ScottyLabs Website leads"
+       id                        = (known after apply)
+       includes_all_repositories = false
+       name                      = "scottylabs-org-leads"
+       organization              = (known after apply)
+       organization_id           = 3
+       permission                = "admin"
+       units_map                 = {
          + "repo.actions"    = "admin"
          + "repo.code"       = "admin"
          + "repo.ext_issues" = "admin"
          + "repo.ext_wiki"   = "admin"
          + "repo.issues"     = "admin"
          + "repo.packages"   = "admin"
          + "repo.projects"   = "admin"
          + "repo.pulls"      = "admin"
          + "repo.releases"   = "admin"
          + "repo.wiki"       = "admin"
        }
    }

  # forgejo_team_member.bus-sign_leads_anish will be created
+ resource "forgejo_team_member" "bus-sign_leads_anish" {
+       team_id = 21
+       user    = "anish"
    }

  # forgejo_team_member.cal_leads_anish will be created
+ resource "forgejo_team_member" "cal_leads_anish" {
+       team_id = 30
+       user    = "anish"
    }

  # forgejo_team_member.cbp_anish will be created
+ resource "forgejo_team_member" "cbp_anish" {
+       team_id = 3
+       user    = "anish"
    }

  # forgejo_team_member.cbp_nikhilr will be created
+ resource "forgejo_team_member" "cbp_nikhilr" {
+       team_id = 3
+       user    = "nramanuj"
    }

  # forgejo_team_member.cbp_xboxbedrock will be created
+ resource "forgejo_team_member" "cbp_xboxbedrock" {
+       team_id = 3
+       user    = "xboxbedrock"
    }

  # forgejo_team_member.cmu-cal_anish will be created
+ resource "forgejo_team_member" "cmu-cal_anish" {
+       team_id = 11
+       user    = "anish"
    }

  # forgejo_team_member.cmu-housing_anish will be created
+ resource "forgejo_team_member" "cmu-housing_anish" {
+       team_id = 2
+       user    = "anish"
    }

  # forgejo_team_member.cmugpt-agent_leads_anish will be created
+ resource "forgejo_team_member" "cmugpt-agent_leads_anish" {
+       team_id = 24
+       user    = "anish"
    }

  # forgejo_team_member.cmugpt-mcp-server_leads_anish will be created
+ resource "forgejo_team_member" "cmugpt-mcp-server_leads_anish" {
+       team_id = 26
+       user    = "anish"
    }

  # forgejo_team_member.cmugpt-sms-surface_leads_anish will be created
+ resource "forgejo_team_member" "cmugpt-sms-surface_leads_anish" {
+       team_id = 25
+       user    = "anish"
    }

  # forgejo_team_member.cmugpt-surface_leads_anish will be created
+ resource "forgejo_team_member" "cmugpt-surface_leads_anish" {
+       team_id = 23
+       user    = "anish"
    }

  # forgejo_team_member.components_leads_anish will be created
+ resource "forgejo_team_member" "components_leads_anish" {
+       team_id = (known after apply)
+       user    = "anish"
    }

  # forgejo_team_member.dalmatian_leads_anish will be created
+ resource "forgejo_team_member" "dalmatian_leads_anish" {
+       team_id = 20
+       user    = "anish"
    }

  # forgejo_team_member.discord-verify_leads_anish will be created
+ resource "forgejo_team_member" "discord-verify_leads_anish" {
+       team_id = 22
+       user    = "anish"
    }

  # forgejo_team_member.groupme-mirror_leads_anish will be created
+ resource "forgejo_team_member" "groupme-mirror_leads_anish" {
+       team_id = 19
+       user    = "anish"
    }

  # forgejo_team_member.housing_leads_anish will be created
+ resource "forgejo_team_member" "housing_leads_anish" {
+       team_id = 31
+       user    = "anish"
    }

  # forgejo_team_member.lost-and-found-v2_leads_anish will be created
+ resource "forgejo_team_member" "lost-and-found-v2_leads_anish" {
+       team_id = (known after apply)
+       user    = "anish"
    }

  # forgejo_team_member.lost-and-found_anish will be created
+ resource "forgejo_team_member" "lost-and-found_anish" {
+       team_id = 5
+       user    = "anish"
    }

  # forgejo_team_member.scottylabs-org_leads_anish will be created
+ resource "forgejo_team_member" "scottylabs-org_leads_anish" {
+       team_id = (known after apply)
+       user    = "anish"
    }

  # forgejo_team_member.slai_anish will be created
+ resource "forgejo_team_member" "slai_anish" {
+       team_id = 8
+       user    = "anish"
    }

  # forgejo_team_member.slai_xboxbedrock will be created
+ resource "forgejo_team_member" "slai_xboxbedrock" {
+       team_id = 8
+       user    = "xboxbedrock"
    }

  # forgejo_team_member.tartan-vote_anish will be created
+ resource "forgejo_team_member" "tartan-vote_anish" {
+       team_id = 6
+       user    = "anish"
    }

  # forgejo_team_member.tartan-vote_leads_anish will be created
+ resource "forgejo_team_member" "tartan-vote_leads_anish" {
+       team_id = 18
+       user    = "anish"
    }

  # forgejo_team_member.tartan-vote_tentype will be created
+ resource "forgejo_team_member" "tartan-vote_tentype" {
+       team_id = 6
+       user    = "maxwen"
    }

  # forgejo_team_member.tartan-vote_xboxbedrock will be created
+ resource "forgejo_team_member" "tartan-vote_xboxbedrock" {
+       team_id = 6
+       user    = "xboxbedrock"
    }

  # forgejo_team_member.uia_anish will be created
+ resource "forgejo_team_member" "uia_anish" {
+       team_id = 9
+       user    = "anish"
    }

  # forgejo_team_repository.components_leads_components will be created
+ resource "forgejo_team_repository" "components_leads_components" {
+       owner      = "ScottyLabs"
+       repository = "components"
+       team_id    = (known after apply)
    }

  # forgejo_team_repository.lost-and-found-v2_leads_lost_and_found_v2 will be created
+ resource "forgejo_team_repository" "lost-and-found-v2_leads_lost_and_found_v2" {
+       owner      = "ScottyLabs"
+       repository = "lost-and-found-v2"
+       team_id    = (known after apply)
    }

  # forgejo_team_repository.scottylabs-org_leads_scottylabs_org will be created
+ resource "forgejo_team_repository" "scottylabs-org_leads_scottylabs_org" {
+       owner      = "ScottyLabs"
+       repository = "scottylabs-org"
+       team_id    = (known after apply)
    }

  # github_repository.slai_cmugpt_agent will be updated in-place
! resource "github_repository" "slai_cmugpt_agent" {
        id                                      = "cmugpt-agent"
        name                                    = "cmugpt-agent"
        # (39 unchanged attributes hidden)

-       template {
-           include_all_branches = false -> null
-           owner                = "ScottyLabs" -> null
-           repository           = "python-template" -> null
        }

        # (1 unchanged block hidden)
    }

  # github_repository.slai_cmugpt_surface will be updated in-place
! resource "github_repository" "slai_cmugpt_surface" {
        id                                      = "cmugpt-surface"
        name                                    = "cmugpt-surface"
        # (39 unchanged attributes hidden)

-       template {
-           include_all_branches = false -> null
-           owner                = "scottylabs-labrador" -> null
-           repository           = "ScottyStack" -> null
        }

        # (1 unchanged block hidden)
    }

  # github_team_membership.cbp_anish will be created
+ resource "github_team_membership" "cbp_anish" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "maintainer"
+       team_id  = "17170099"
+       username = "ap-1"
    }

  # github_team_membership.cbp_nikhilr will be created
+ resource "github_team_membership" "cbp_nikhilr" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "17170099"
+       username = "ecstaticpilot"
    }

  # github_team_membership.cbp_xboxbedrock will be created
+ resource "github_team_membership" "cbp_xboxbedrock" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "17170099"
+       username = "xboxbedrock"
    }

  # github_team_membership.cmu-cal_anish will be created
+ resource "github_team_membership" "cmu-cal_anish" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "maintainer"
+       team_id  = "15669112"
+       username = "ap-1"
    }

  # github_team_membership.cmu-housing_anish will be created
+ resource "github_team_membership" "cmu-housing_anish" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "maintainer"
+       team_id  = "17937919"
+       username = "ap-1"
    }

  # github_team_membership.lost-and-found_anish will be created
+ resource "github_team_membership" "lost-and-found_anish" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "maintainer"
+       team_id  = "4088608"
+       username = "ap-1"
    }

  # github_team_membership.slai_anish will be created
+ resource "github_team_membership" "slai_anish" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "maintainer"
+       team_id  = "17571328"
+       username = "ap-1"
    }

  # github_team_membership.slai_xboxbedrock will be created
+ resource "github_team_membership" "slai_xboxbedrock" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "17571328"
+       username = "xboxbedrock"
    }

  # github_team_membership.tartan-vote_anish will be created
+ resource "github_team_membership" "tartan-vote_anish" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "maintainer"
+       team_id  = "17701833"
+       username = "ap-1"
    }

  # github_team_membership.tartan-vote_tentype will be created
+ resource "github_team_membership" "tartan-vote_tentype" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "17701833"
+       username = "tentype"
    }

  # github_team_membership.tartan-vote_xboxbedrock will be created
+ resource "github_team_membership" "tartan-vote_xboxbedrock" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "17701833"
+       username = "xboxbedrock"
    }

  # github_team_membership.uia_anish will be created
+ resource "github_team_membership" "uia_anish" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "maintainer"
+       team_id  = "17529763"
+       username = "ap-1"
    }

Plan: 44 to add, 2 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 44 to add, 2 to change, 0 to destroy.


Plan Summary

5 projects, 5 with changes, 0 with no changes, 0 failed

  • To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 5 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Compiling proc-macro2 v1.0.106 Compiling quote v1.0.45 Compiling unicode-ident v1.0.24 Compiling serde_core v1.0.228 Compiling shlex v1.3.0 Compiling find-msvc-tools v0.1.9 Compiling libc v0.2.183 Compiling cc v1.2.57 Compiling cfg-if v1.0.4 Compiling zmij v1.0.21 Compiling syn v2.0.117 Compiling itoa v1.0.18 Compiling ring v0.17.14 Compiling zeroize v1.8.2 Compiling serde_json v1.0.149 Compiling serde v1.0.228 Compiling rustls-pki-types v1.14.0 Compiling getrandom v0.2.17 Compiling ref-cast v1.0.25 Compiling untrusted v0.9.0 Compiling crc32fast v1.5.0 Compiling httparse v1.10.1 Compiling memchr v2.8.0 Compiling serde_derive_internals v0.29.1 Compiling serde_derive v1.0.228 Compiling ref-cast-impl v1.0.25 Compiling simd-adler32 v0.3.8 Compiling bytes v1.11.1 Compiling rustls v0.23.37 Compiling adler2 v2.0.1 Compiling log v0.4.29 Compiling miniz_oxide v0.8.9 Compiling http v1.4.0 Compiling schemars_derive v1.2.1 Compiling rustls-webpki v0.103.13 Compiling thiserror v2.0.18 Compiling dyn-clone v1.0.20 Compiling base64 v0.22.1 Compiling once_cell v1.21.4 Compiling winnow v1.0.0 Compiling subtle v2.6.1 Compiling utf8parse v0.2.2 Compiling anstyle-parse v1.0.0 Compiling toml_parser v1.1.0+spec-1.1.0 Compiling ureq-proto v0.6.0 Compiling schemars v1.2.1 Compiling flate2 v1.1.9 Compiling thiserror-impl v2.0.18 Compiling webpki-roots v1.0.6 Compiling serde_spanned v1.1.0 Compiling toml_datetime v1.1.0+spec-1.1.0 Compiling is_terminal_polyfill v1.70.2 Compiling utf8-zero v0.8.1 Compiling anstyle v1.0.14 Compiling toml_writer v1.1.0+spec-1.1.0 Compiling percent-encoding v2.3.2 Compiling anstyle-query v1.1.5 Compiling colorchoice v1.0.5 Compiling anstream v1.0.0 Compiling toml v1.1.0+spec-1.1.0 Compiling ureq v3.3.0 Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/194/default/crates/governance-schema) Compiling clap_lex v1.1.0 Compiling anyhow v1.0.104 Compiling heck v0.5.0 Compiling strsim v0.11.1 Compiling clap_builder v4.6.0 Compiling clap_derive v4.6.0 Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/194/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/194/default/crates/governance-tfgen) Compiling clap v4.6.0 Compiling dotenvy v0.15.7 Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/194/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 45.60s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_cbp_ch0_anish will be created + resource "null_resource" "slack_cbp_ch0_anish" { + id = (known after apply) + triggers = { + "channel" = "C08S56CJ8R2" + "user" = "U07FPJKFB5E" } } # null_resource.slack_cbp_ch0_nikhilr will be created + resource "null_resource" "slack_cbp_ch0_nikhilr" { + id = (known after apply) + triggers = { + "channel" = "C08S56CJ8R2" + "user" = "U0AMBE2QKBM" } } # null_resource.slack_cbp_ch0_xboxbedrock will be created + resource "null_resource" "slack_cbp_ch0_xboxbedrock" { + id = (known after apply) + triggers = { + "channel" = "C08S56CJ8R2" + "user" = "U0AQRKYLGHK" } } # null_resource.slack_cmu-cal_ch0_anish will be created + resource "null_resource" "slack_cmu-cal_ch0_anish" { + id = (known after apply) + triggers = { + "channel" = "C05TNBPHVTL" + "user" = "U07FPJKFB5E" } } # null_resource.slack_lost-and-found_ch0_anish will be created + resource "null_resource" "slack_lost-and-found_ch0_anish" { + id = (known after apply) + triggers = { + "channel" = "C01BAADS8U8" + "user" = "U07FPJKFB5E" } } # null_resource.slack_slai_ch0_anish will be created + resource "null_resource" "slack_slai_ch0_anish" { + id = (known after apply) + triggers = { + "channel" = "C08SRUGUQ05" + "user" = "U07FPJKFB5E" } } # null_resource.slack_slai_ch0_xboxbedrock will be created + resource "null_resource" "slack_slai_ch0_xboxbedrock" { + id = (known after apply) + triggers = { + "channel" = "C08SRUGUQ05" + "user" = "U0AQRKYLGHK" } } # null_resource.slack_uia_ch0_anish will be created + resource "null_resource" "slack_uia_ch0_anish" { + id = (known after apply) + triggers = { + "channel" = "C08MAT652CE" + "user" = "U07FPJKFB5E" } } Plan: 8 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 8 to add, 0 to change, 0 to destroy. --- ### 2. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 45.66s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_bridge_login_org_labrador will be created + resource "null_resource" "slack_bridge_login_org_labrador" { + id = (known after apply) + triggers = { + "channel" = "C05TG63EM9B" } } # null_resource.slack_bridge_login_org_outreach will be created + resource "null_resource" "slack_bridge_login_org_outreach" { + id = (known after apply) + triggers = { + "channel" = "C041U79AZAT" } } # synapse_mautrix_slack_link.org_labrador will be created + resource "synapse_mautrix_slack_link" "org_labrador" { + discord_channel_id = "1549280062581252187" + discord_relay_configured = (known after apply) + id = (known after apply) + project_name = "Labrador" + project_slug = "labrador" + room_alias = (known after apply) + room_id = (known after apply) + slack_bridged = (known after apply) + slack_channel_id = "C05TG63EM9B" + team_name = "ScottyLabs" + team_slug = "org" } # synapse_mautrix_slack_link.org_outreach will be created + resource "synapse_mautrix_slack_link" "org_outreach" { + discord_channel_id = "1549280019686101182" + discord_relay_configured = (known after apply) + id = (known after apply) + project_name = "Outreach" + project_slug = "outreach" + room_alias = (known after apply) + room_id = (known after apply) + slack_bridged = (known after apply) + slack_channel_id = "C041U79AZAT" + team_name = "ScottyLabs" + team_slug = "org" } Plan: 4 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p matrix_bridges ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fmatrix_bridges%252Fdefault%252Fmatrix_bridges) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p matrix_bridges ``` Plan: 4 to add, 0 to change, 0 to destroy. --- ### 3. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 45.62s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cbp_bus-sign will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_bus-sign" { id = "scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b" ! members = [ + "apallati", + "nramanuj", + "spola2", # (3 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_bus-sign_admins will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_bus-sign_admins" { id = "scottylabs/group-memberships/100e7090-c4a4-47f2-b0ad-8eb81dd8c8c8" ! members = [ + "apallati", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_dalmatian will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_dalmatian" { id = "scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153" ! members = [ + "apallati", + "nramanuj", + "spola2", # (3 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_dalmatian_admins will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_dalmatian_admins" { id = "scottylabs/group-memberships/5467214e-843c-488a-85fb-2ca5fa77d7d7" ! members = [ + "apallati", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_discord-verify will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_discord-verify" { id = "scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745" ! members = [ + "apallati", + "nramanuj", + "spola2", # (3 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_discord-verify_admins will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_discord-verify_admins" { id = "scottylabs/group-memberships/a2abc505-3248-42f8-9513-d3db72be5d40" ! members = [ + "apallati", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_groupme-mirror will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_groupme-mirror" { id = "scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39" ! members = [ + "apallati", + "nramanuj", + "spola2", # (3 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cbp_groupme-mirror_admins will be updated in-place ! resource "keycloak_group_memberships" "project_cbp_groupme-mirror_admins" { id = "scottylabs/group-memberships/c4c2efc2-d322-4193-8279-699e397522b0" ! members = [ + "apallati", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-cal_cal will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-cal_cal" { id = "scottylabs/group-memberships/3800a836-0a57-40b7-839f-650a335bfccc" ! members = [ + "apallati", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-cal_cal_admins will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-cal_cal_admins" { id = "scottylabs/group-memberships/5ca15cf1-036f-4644-b45e-8ca25ce9b2af" ! members = [ + "apallati", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-housing_housing" { id = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d" ! members = [ + "apallati", # (10 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_cmu-housing_housing_admins will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-housing_housing_admins" { id = "scottylabs/group-memberships/7c932acc-ac1e-420b-8040-9c3af456b597" ! members = [ + "apallati", # (4 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_lost-and-found_lost-and-found-v2 will be updated in-place ! resource "keycloak_group_memberships" "project_lost-and-found_lost-and-found-v2" { id = "scottylabs/group-memberships/c14b4061-e3db-42c9-9b9e-a0c0a9747ce4" ! members = [ + "apallati", ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_lost-and-found_lost-and-found-v2_admins will be updated in-place ! resource "keycloak_group_memberships" "project_lost-and-found_lost-and-found-v2_admins" { id = "scottylabs/group-memberships/7879878f-16a8-4b07-835e-81a8a9e7b02f" ! members = [ + "apallati", ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-agent will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-agent" { id = "scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423" ! members = [ + "apallati", + "spola2", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-agent_admins will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-agent_admins" { id = "scottylabs/group-memberships/f4e82f9b-769d-4f38-a7f2-e214189d9269" ! members = [ + "apallati", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-mcp-server will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server" { id = "scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8" ! members = [ + "apallati", + "spola2", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-mcp-server_admins will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-mcp-server_admins" { id = "scottylabs/group-memberships/8f84f9ec-e78a-41f6-9ce6-097fce273298" ! members = [ + "apallati", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-sms-surface will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface" { id = "scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43" ! members = [ + "apallati", + "spola2", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-sms-surface_admins will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-sms-surface_admins" { id = "scottylabs/group-memberships/34404457-985f-4a61-b2f8-4754c2eb3293" ! members = [ + "apallati", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-surface will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-surface" { id = "scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775" ! members = [ + "apallati", + "spola2", # (5 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_slai_cmugpt-surface_admins will be updated in-place ! resource "keycloak_group_memberships" "project_slai_cmugpt-surface_admins" { id = "scottylabs/group-memberships/4f3531db-8075-480e-bbf4-253e98783b56" ! members = [ + "apallati", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_tartan-vote will be updated in-place ! resource "keycloak_group_memberships" "project_tartan-vote" { id = "scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c" ! members = [ + "apallati", + "maxwen", + "spola2", # (2 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_tartan-vote_admins will be updated in-place ! resource "keycloak_group_memberships" "project_tartan-vote_admins" { id = "scottylabs/group-memberships/cee98bbd-fb26-4527-ac5d-8a8587de2416" ! members = [ + "apallati", # (1 unchanged element hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_uia_components will be updated in-place ! resource "keycloak_group_memberships" "project_uia_components" { id = "scottylabs/group-memberships/95465a93-b14a-459d-85b5-12f8b0fe7fd2" ! members = [ + "apallati", ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_uia_components_admins will be updated in-place ! resource "keycloak_group_memberships" "project_uia_components_admins" { id = "scottylabs/group-memberships/75073d65-02c5-405b-bc54-b4036dbda934" ! members = [ + "apallati", ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_uia_scottylabs-org will be updated in-place ! resource "keycloak_group_memberships" "project_uia_scottylabs-org" { id = "scottylabs/group-memberships/f7ea42f9-6007-4a5d-b46b-e5ac1d3e2bb8" ! members = [ + "apallati", ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.project_uia_scottylabs-org_admins will be updated in-place ! resource "keycloak_group_memberships" "project_uia_scottylabs-org_admins" { id = "scottylabs/group-memberships/5786d2ba-5816-4ce2-8f04-d76958c4ee91" ! members = [ + "apallati", ] # (2 unchanged attributes hidden) } Plan: 0 to add, 28 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 28 to change, 0 to destroy. --- ### 4. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 45.61s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # discord_member_roles.anish will be updated in-place ! resource "discord_member_roles" "anish" { id = "1353118209024331948:315249559473225729" # (2 unchanged attributes hidden) + role { + has_role = true + role_id = "1494773174334783499" } + role { + has_role = true + role_id = "1504199977000505445" } + role { + has_role = true + role_id = "1505078357321191535" } + role { + has_role = true + role_id = "1508601984255524885" } + role { + has_role = true + role_id = "1513653499957874871" } + role { + has_role = true + role_id = "1526500755446698124" } + role { + has_role = true + role_id = "1526511044762271914" } # (9 unchanged blocks hidden) } # discord_member_roles.nikhilr will be updated in-place ! resource "discord_member_roles" "nikhilr" { id = "1353118209024331948:795153900553764874" # (2 unchanged attributes hidden) + role { + has_role = true + role_id = "1494773174334783499" } # (3 unchanged blocks hidden) } # discord_member_roles.tentype will be updated in-place ! resource "discord_member_roles" "tentype" { id = "1353118209024331948:637528457508552705" # (2 unchanged attributes hidden) + role { + has_role = true + role_id = "1508601984255524885" } # (4 unchanged blocks hidden) } # discord_member_roles.xboxbedrock will be updated in-place ! resource "discord_member_roles" "xboxbedrock" { id = "1353118209024331948:712398210345205871" # (2 unchanged attributes hidden) + role { + has_role = true + role_id = "1494773174334783499" } + role { + has_role = true + role_id = "1505078357321191535" } + role { + has_role = true + role_id = "1508601984255524885" } # (6 unchanged blocks hidden) } Plan: 0 to add, 4 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 0 to add, 4 to change, 0 to destroy. --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 45.65s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # forgejo_team.components_leads will be created + resource "forgejo_team" "components_leads" { + can_create_org_repo = false + description = "Components leads" + id = (known after apply) + includes_all_repositories = false + name = "components-leads" + organization = (known after apply) + organization_id = 3 + permission = "admin" + units_map = { + "repo.actions" = "admin" + "repo.code" = "admin" + "repo.ext_issues" = "admin" + "repo.ext_wiki" = "admin" + "repo.issues" = "admin" + "repo.packages" = "admin" + "repo.projects" = "admin" + "repo.pulls" = "admin" + "repo.releases" = "admin" + "repo.wiki" = "admin" } } # forgejo_team.lost-and-found-v2_leads will be created + resource "forgejo_team" "lost-and-found-v2_leads" { + can_create_org_repo = false + description = "Lost and Found leads" + id = (known after apply) + includes_all_repositories = false + name = "lost-and-found-v2-leads" + organization = (known after apply) + organization_id = 3 + permission = "admin" + units_map = { + "repo.actions" = "admin" + "repo.code" = "admin" + "repo.ext_issues" = "admin" + "repo.ext_wiki" = "admin" + "repo.issues" = "admin" + "repo.packages" = "admin" + "repo.projects" = "admin" + "repo.pulls" = "admin" + "repo.releases" = "admin" + "repo.wiki" = "admin" } } # forgejo_team.scottylabs-org_leads will be created + resource "forgejo_team" "scottylabs-org_leads" { + can_create_org_repo = false + description = "ScottyLabs Website leads" + id = (known after apply) + includes_all_repositories = false + name = "scottylabs-org-leads" + organization = (known after apply) + organization_id = 3 + permission = "admin" + units_map = { + "repo.actions" = "admin" + "repo.code" = "admin" + "repo.ext_issues" = "admin" + "repo.ext_wiki" = "admin" + "repo.issues" = "admin" + "repo.packages" = "admin" + "repo.projects" = "admin" + "repo.pulls" = "admin" + "repo.releases" = "admin" + "repo.wiki" = "admin" } } # forgejo_team_member.bus-sign_leads_anish will be created + resource "forgejo_team_member" "bus-sign_leads_anish" { + team_id = 21 + user = "anish" } # forgejo_team_member.cal_leads_anish will be created + resource "forgejo_team_member" "cal_leads_anish" { + team_id = 30 + user = "anish" } # forgejo_team_member.cbp_anish will be created + resource "forgejo_team_member" "cbp_anish" { + team_id = 3 + user = "anish" } # forgejo_team_member.cbp_nikhilr will be created + resource "forgejo_team_member" "cbp_nikhilr" { + team_id = 3 + user = "nramanuj" } # forgejo_team_member.cbp_xboxbedrock will be created + resource "forgejo_team_member" "cbp_xboxbedrock" { + team_id = 3 + user = "xboxbedrock" } # forgejo_team_member.cmu-cal_anish will be created + resource "forgejo_team_member" "cmu-cal_anish" { + team_id = 11 + user = "anish" } # forgejo_team_member.cmu-housing_anish will be created + resource "forgejo_team_member" "cmu-housing_anish" { + team_id = 2 + user = "anish" } # forgejo_team_member.cmugpt-agent_leads_anish will be created + resource "forgejo_team_member" "cmugpt-agent_leads_anish" { + team_id = 24 + user = "anish" } # forgejo_team_member.cmugpt-mcp-server_leads_anish will be created + resource "forgejo_team_member" "cmugpt-mcp-server_leads_anish" { + team_id = 26 + user = "anish" } # forgejo_team_member.cmugpt-sms-surface_leads_anish will be created + resource "forgejo_team_member" "cmugpt-sms-surface_leads_anish" { + team_id = 25 + user = "anish" } # forgejo_team_member.cmugpt-surface_leads_anish will be created + resource "forgejo_team_member" "cmugpt-surface_leads_anish" { + team_id = 23 + user = "anish" } # forgejo_team_member.components_leads_anish will be created + resource "forgejo_team_member" "components_leads_anish" { + team_id = (known after apply) + user = "anish" } # forgejo_team_member.dalmatian_leads_anish will be created + resource "forgejo_team_member" "dalmatian_leads_anish" { + team_id = 20 + user = "anish" } # forgejo_team_member.discord-verify_leads_anish will be created + resource "forgejo_team_member" "discord-verify_leads_anish" { + team_id = 22 + user = "anish" } # forgejo_team_member.groupme-mirror_leads_anish will be created + resource "forgejo_team_member" "groupme-mirror_leads_anish" { + team_id = 19 + user = "anish" } # forgejo_team_member.housing_leads_anish will be created + resource "forgejo_team_member" "housing_leads_anish" { + team_id = 31 + user = "anish" } # forgejo_team_member.lost-and-found-v2_leads_anish will be created + resource "forgejo_team_member" "lost-and-found-v2_leads_anish" { + team_id = (known after apply) + user = "anish" } # forgejo_team_member.lost-and-found_anish will be created + resource "forgejo_team_member" "lost-and-found_anish" { + team_id = 5 + user = "anish" } # forgejo_team_member.scottylabs-org_leads_anish will be created + resource "forgejo_team_member" "scottylabs-org_leads_anish" { + team_id = (known after apply) + user = "anish" } # forgejo_team_member.slai_anish will be created + resource "forgejo_team_member" "slai_anish" { + team_id = 8 + user = "anish" } # forgejo_team_member.slai_xboxbedrock will be created + resource "forgejo_team_member" "slai_xboxbedrock" { + team_id = 8 + user = "xboxbedrock" } # forgejo_team_member.tartan-vote_anish will be created + resource "forgejo_team_member" "tartan-vote_anish" { + team_id = 6 + user = "anish" } # forgejo_team_member.tartan-vote_leads_anish will be created + resource "forgejo_team_member" "tartan-vote_leads_anish" { + team_id = 18 + user = "anish" } # forgejo_team_member.tartan-vote_tentype will be created + resource "forgejo_team_member" "tartan-vote_tentype" { + team_id = 6 + user = "maxwen" } # forgejo_team_member.tartan-vote_xboxbedrock will be created + resource "forgejo_team_member" "tartan-vote_xboxbedrock" { + team_id = 6 + user = "xboxbedrock" } # forgejo_team_member.uia_anish will be created + resource "forgejo_team_member" "uia_anish" { + team_id = 9 + user = "anish" } # forgejo_team_repository.components_leads_components will be created + resource "forgejo_team_repository" "components_leads_components" { + owner = "ScottyLabs" + repository = "components" + team_id = (known after apply) } # forgejo_team_repository.lost-and-found-v2_leads_lost_and_found_v2 will be created + resource "forgejo_team_repository" "lost-and-found-v2_leads_lost_and_found_v2" { + owner = "ScottyLabs" + repository = "lost-and-found-v2" + team_id = (known after apply) } # forgejo_team_repository.scottylabs-org_leads_scottylabs_org will be created + resource "forgejo_team_repository" "scottylabs-org_leads_scottylabs_org" { + owner = "ScottyLabs" + repository = "scottylabs-org" + team_id = (known after apply) } # github_repository.slai_cmugpt_agent will be updated in-place ! resource "github_repository" "slai_cmugpt_agent" { id = "cmugpt-agent" name = "cmugpt-agent" # (39 unchanged attributes hidden) - template { - include_all_branches = false -> null - owner = "ScottyLabs" -> null - repository = "python-template" -> null } # (1 unchanged block hidden) } # github_repository.slai_cmugpt_surface will be updated in-place ! resource "github_repository" "slai_cmugpt_surface" { id = "cmugpt-surface" name = "cmugpt-surface" # (39 unchanged attributes hidden) - template { - include_all_branches = false -> null - owner = "scottylabs-labrador" -> null - repository = "ScottyStack" -> null } # (1 unchanged block hidden) } # github_team_membership.cbp_anish will be created + resource "github_team_membership" "cbp_anish" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = "17170099" + username = "ap-1" } # github_team_membership.cbp_nikhilr will be created + resource "github_team_membership" "cbp_nikhilr" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "17170099" + username = "ecstaticpilot" } # github_team_membership.cbp_xboxbedrock will be created + resource "github_team_membership" "cbp_xboxbedrock" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "17170099" + username = "xboxbedrock" } # github_team_membership.cmu-cal_anish will be created + resource "github_team_membership" "cmu-cal_anish" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = "15669112" + username = "ap-1" } # github_team_membership.cmu-housing_anish will be created + resource "github_team_membership" "cmu-housing_anish" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = "17937919" + username = "ap-1" } # github_team_membership.lost-and-found_anish will be created + resource "github_team_membership" "lost-and-found_anish" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = "4088608" + username = "ap-1" } # github_team_membership.slai_anish will be created + resource "github_team_membership" "slai_anish" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = "17571328" + username = "ap-1" } # github_team_membership.slai_xboxbedrock will be created + resource "github_team_membership" "slai_xboxbedrock" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "17571328" + username = "xboxbedrock" } # github_team_membership.tartan-vote_anish will be created + resource "github_team_membership" "tartan-vote_anish" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = "17701833" + username = "ap-1" } # github_team_membership.tartan-vote_tentype will be created + resource "github_team_membership" "tartan-vote_tentype" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "17701833" + username = "tentype" } # github_team_membership.tartan-vote_xboxbedrock will be created + resource "github_team_membership" "tartan-vote_xboxbedrock" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "17701833" + username = "xboxbedrock" } # github_team_membership.uia_anish will be created + resource "github_team_membership" "uia_anish" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = "17529763" + username = "ap-1" } Plan: 44 to add, 2 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 44 to add, 2 to change, 0 to destroy. --- ### Plan Summary 5 projects, 5 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Author
Owner

atlantis apply

atlantis apply

Ran Apply for 5 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: keycloak dir: tofu/keycloak workspace: default
  4. project: matrix_bridges dir: tofu/matrix_bridges workspace: default
  5. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.19s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_cbp_ch0_nikhilr: Creating...
null_resource.slack_cbp_ch0_anish: Creating...
null_resource.slack_lost-and-found_ch0_anish: Creating...
null_resource.slack_lost-and-found_ch0_anish: Provisioning with 'local-exec'...
null_resource.slack_lost-and-found_ch0_anish (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C01BAADS8U8 --user U07FPJKFB5E"]
null_resource.slack_slai_ch0_anish: Creating...
null_resource.slack_cbp_ch0_xboxbedrock: Creating...
null_resource.slack_slai_ch0_xboxbedrock: Creating...
null_resource.slack_cbp_ch0_nikhilr: Provisioning with 'local-exec'...
null_resource.slack_cbp_ch0_nikhilr (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08S56CJ8R2 --user U0AMBE2QKBM"]
null_resource.slack_cbp_ch0_anish: Provisioning with 'local-exec'...
null_resource.slack_cbp_ch0_anish (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08S56CJ8R2 --user U07FPJKFB5E"]
null_resource.slack_cbp_ch0_xboxbedrock: Provisioning with 'local-exec'...
null_resource.slack_cbp_ch0_xboxbedrock (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08S56CJ8R2 --user U0AQRKYLGHK"]
null_resource.slack_uia_ch0_anish: Creating...
null_resource.slack_cmu-cal_ch0_anish: Creating...
null_resource.slack_slai_ch0_anish: Provisioning with 'local-exec'...
null_resource.slack_slai_ch0_anish (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08SRUGUQ05 --user U07FPJKFB5E"]
null_resource.slack_slai_ch0_xboxbedrock: Provisioning with 'local-exec'...
null_resource.slack_slai_ch0_xboxbedrock (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08SRUGUQ05 --user U0AQRKYLGHK"]
null_resource.slack_uia_ch0_anish: Provisioning with 'local-exec'...
null_resource.slack_uia_ch0_anish (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08MAT652CE --user U07FPJKFB5E"]
null_resource.slack_cmu-cal_ch0_anish: Provisioning with 'local-exec'...
null_resource.slack_cmu-cal_ch0_anish (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C05TNBPHVTL --user U07FPJKFB5E"]
null_resource.slack_slai_ch0_anish: Creation complete after 0s [id=5303212720340085892]
null_resource.slack_slai_ch0_xboxbedrock: Creation complete after 0s [id=7816746119839820188]
null_resource.slack_uia_ch0_anish: Creation complete after 0s [id=4805975139938897672]
null_resource.slack_cmu-cal_ch0_anish: Creation complete after 0s [id=3978892587047122247]
null_resource.slack_cbp_ch0_anish: Creation complete after 0s [id=2983489810514487445]
null_resource.slack_cbp_ch0_nikhilr: Creation complete after 0s [id=4626205003162193013]
null_resource.slack_lost-and-found_ch0_anish: Creation complete after 0s [id=5751452929757186483]
null_resource.slack_cbp_ch0_xboxbedrock: Creation complete after 1s [id=1848007231579617513]

Apply complete! Resources: 8 added, 0 changed, 0 destroyed.

2. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.20s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.nikhilr: Modifying... [id=1353118209024331948:795153900553764874]
discord_member_roles.tentype: Modifying... [id=1353118209024331948:637528457508552705]
discord_member_roles.anish: Modifying... [id=1353118209024331948:315249559473225729]
discord_member_roles.xboxbedrock: Modifying... [id=1353118209024331948:712398210345205871]
discord_member_roles.nikhilr: Modifications complete after 1s [id=1353118209024331948:795153900553764874]
discord_member_roles.tentype: Modifications complete after 1s [id=1353118209024331948:637528457508552705]
discord_member_roles.anish: Modifications complete after 2s [id=1353118209024331948:315249559473225729]
discord_member_roles.xboxbedrock: Modifications complete after 2s [id=1353118209024331948:712398210345205871]

Apply complete! Resources: 0 added, 4 changed, 0 destroyed.

3. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.21s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cbp_dalmatian_admins: Modifying... [id=scottylabs/group-memberships/5467214e-843c-488a-85fb-2ca5fa77d7d7]
keycloak_group_memberships.project_cbp_dalmatian: Modifying... [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153]
keycloak_group_memberships.project_uia_components: Modifying... [id=scottylabs/group-memberships/95465a93-b14a-459d-85b5-12f8b0fe7fd2]
keycloak_group_memberships.project_slai_cmugpt-sms-surface: Modifying... [id=scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43]
keycloak_group_memberships.project_lost-and-found_lost-and-found-v2_admins: Modifying... [id=scottylabs/group-memberships/7879878f-16a8-4b07-835e-81a8a9e7b02f]
keycloak_group_memberships.project_tartan-vote_admins: Modifying... [id=scottylabs/group-memberships/cee98bbd-fb26-4527-ac5d-8a8587de2416]
keycloak_group_memberships.project_slai_cmugpt-agent: Modifying... [id=scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423]
keycloak_group_memberships.project_uia_components_admins: Modifying... [id=scottylabs/group-memberships/75073d65-02c5-405b-bc54-b4036dbda934]
keycloak_group_memberships.project_slai_cmugpt-mcp-server_admins: Modifying... [id=scottylabs/group-memberships/8f84f9ec-e78a-41f6-9ce6-097fce273298]
keycloak_group_memberships.project_slai_cmugpt-agent_admins: Modifying... [id=scottylabs/group-memberships/f4e82f9b-769d-4f38-a7f2-e214189d9269]
keycloak_group_memberships.project_slai_cmugpt-agent_admins: Modifications complete after 2s [id=scottylabs/group-memberships/f4e82f9b-769d-4f38-a7f2-e214189d9269]
keycloak_group_memberships.project_cmu-cal_cal: Modifying... [id=scottylabs/group-memberships/3800a836-0a57-40b7-839f-650a335bfccc]
keycloak_group_memberships.project_cbp_dalmatian_admins: Modifications complete after 2s [id=scottylabs/group-memberships/5467214e-843c-488a-85fb-2ca5fa77d7d7]
keycloak_group_memberships.project_slai_cmugpt-mcp-server: Modifying... [id=scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8]
keycloak_group_memberships.project_uia_components_admins: Modifications complete after 2s [id=scottylabs/group-memberships/75073d65-02c5-405b-bc54-b4036dbda934]
keycloak_group_memberships.project_tartan-vote: Modifying... [id=scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c]
keycloak_group_memberships.project_slai_cmugpt-mcp-server_admins: Modifications complete after 2s [id=scottylabs/group-memberships/8f84f9ec-e78a-41f6-9ce6-097fce273298]
keycloak_group_memberships.project_slai_cmugpt-sms-surface_admins: Modifying... [id=scottylabs/group-memberships/34404457-985f-4a61-b2f8-4754c2eb3293]
keycloak_group_memberships.project_uia_components: Modifications complete after 2s [id=scottylabs/group-memberships/95465a93-b14a-459d-85b5-12f8b0fe7fd2]
keycloak_group_memberships.project_cbp_discord-verify: Modifying... [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745]
keycloak_group_memberships.project_lost-and-found_lost-and-found-v2_admins: Modifications complete after 2s [id=scottylabs/group-memberships/7879878f-16a8-4b07-835e-81a8a9e7b02f]
keycloak_group_memberships.project_cmu-cal_cal_admins: Modifying... [id=scottylabs/group-memberships/5ca15cf1-036f-4644-b45e-8ca25ce9b2af]
keycloak_group_memberships.project_tartan-vote_admins: Modifications complete after 2s [id=scottylabs/group-memberships/cee98bbd-fb26-4527-ac5d-8a8587de2416]
keycloak_group_memberships.project_cbp_discord-verify_admins: Modifying... [id=scottylabs/group-memberships/a2abc505-3248-42f8-9513-d3db72be5d40]
keycloak_group_memberships.project_slai_cmugpt-agent: Modifications complete after 4s [id=scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423]
keycloak_group_memberships.project_lost-and-found_lost-and-found-v2: Modifying... [id=scottylabs/group-memberships/c14b4061-e3db-42c9-9b9e-a0c0a9747ce4]
keycloak_group_memberships.project_slai_cmugpt-sms-surface: Modifications complete after 4s [id=scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43]
keycloak_group_memberships.project_uia_scottylabs-org: Modifying... [id=scottylabs/group-memberships/f7ea42f9-6007-4a5d-b46b-e5ac1d3e2bb8]
keycloak_group_memberships.project_cmu-cal_cal: Modifications complete after 2s [id=scottylabs/group-memberships/3800a836-0a57-40b7-839f-650a335bfccc]
keycloak_group_memberships.project_uia_scottylabs-org_admins: Modifying... [id=scottylabs/group-memberships/5786d2ba-5816-4ce2-8f04-d76958c4ee91]
keycloak_group_memberships.project_cmu-cal_cal_admins: Modifications complete after 2s [id=scottylabs/group-memberships/5ca15cf1-036f-4644-b45e-8ca25ce9b2af]
keycloak_group_memberships.project_cmu-housing_housing: Modifying... [id=scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d]
keycloak_group_memberships.project_slai_cmugpt-sms-surface_admins: Modifications complete after 2s [id=scottylabs/group-memberships/34404457-985f-4a61-b2f8-4754c2eb3293]
keycloak_group_memberships.project_cbp_groupme-mirror: Modifying... [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39]
keycloak_group_memberships.project_cbp_discord-verify_admins: Modifications complete after 2s [id=scottylabs/group-memberships/a2abc505-3248-42f8-9513-d3db72be5d40]
keycloak_group_memberships.project_slai_cmugpt-surface: Modifying... [id=scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775]
keycloak_group_memberships.project_slai_cmugpt-mcp-server: Modifications complete after 4s [id=scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8]
keycloak_group_memberships.project_cbp_bus-sign: Modifying... [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b]
keycloak_group_memberships.project_uia_scottylabs-org: Modifications complete after 2s [id=scottylabs/group-memberships/f7ea42f9-6007-4a5d-b46b-e5ac1d3e2bb8]
keycloak_group_memberships.project_cmu-housing_housing_admins: Modifying... [id=scottylabs/group-memberships/7c932acc-ac1e-420b-8040-9c3af456b597]
keycloak_group_memberships.project_lost-and-found_lost-and-found-v2: Modifications complete after 2s [id=scottylabs/group-memberships/c14b4061-e3db-42c9-9b9e-a0c0a9747ce4]
keycloak_group_memberships.project_cbp_groupme-mirror_admins: Modifying... [id=scottylabs/group-memberships/c4c2efc2-d322-4193-8279-699e397522b0]
keycloak_group_memberships.project_uia_scottylabs-org_admins: Modifications complete after 3s [id=scottylabs/group-memberships/5786d2ba-5816-4ce2-8f04-d76958c4ee91]
keycloak_group_memberships.project_slai_cmugpt-surface_admins: Modifying... [id=scottylabs/group-memberships/4f3531db-8075-480e-bbf4-253e98783b56]
keycloak_group_memberships.project_cbp_dalmatian: Modifications complete after 7s [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153]
keycloak_group_memberships.project_cbp_bus-sign_admins: Modifying... [id=scottylabs/group-memberships/100e7090-c4a4-47f2-b0ad-8eb81dd8c8c8]
keycloak_group_memberships.project_cmu-housing_housing: Modifications complete after 3s [id=scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d]
keycloak_group_memberships.project_tartan-vote: Modifications complete after 6s [id=scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c]
keycloak_group_memberships.project_cmu-housing_housing_admins: Modifications complete after 2s [id=scottylabs/group-memberships/7c932acc-ac1e-420b-8040-9c3af456b597]
keycloak_group_memberships.project_slai_cmugpt-surface: Modifications complete after 4s [id=scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775]
keycloak_group_memberships.project_slai_cmugpt-surface_admins: Modifications complete after 2s [id=scottylabs/group-memberships/4f3531db-8075-480e-bbf4-253e98783b56]
keycloak_group_memberships.project_cbp_groupme-mirror_admins: Modifications complete after 3s [id=scottylabs/group-memberships/c4c2efc2-d322-4193-8279-699e397522b0]
keycloak_group_memberships.project_cbp_discord-verify: Modifications complete after 7s [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745]
keycloak_group_memberships.project_cbp_bus-sign_admins: Modifications complete after 2s [id=scottylabs/group-memberships/100e7090-c4a4-47f2-b0ad-8eb81dd8c8c8]
keycloak_group_memberships.project_cbp_groupme-mirror: Modifications complete after 6s [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39]
keycloak_group_memberships.project_cbp_bus-sign: Modifications complete after 6s [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b]

Apply complete! Resources: 0 added, 28 changed, 0 destroyed.

4. project: matrix_bridges dir: tofu/matrix_bridges workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.12s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_bridge_login_org_labrador: Creating...
null_resource.slack_bridge_login_org_labrador: Provisioning with 'local-exec'...
null_resource.slack_bridge_login_org_labrador (local-exec): Executing: ["/bin/sh" "-c" "governance slack-join --channel C05TG63EM9B"]
null_resource.slack_bridge_login_org_outreach: Creating...
null_resource.slack_bridge_login_org_outreach: Provisioning with 'local-exec'...
null_resource.slack_bridge_login_org_outreach (local-exec): Executing: ["/bin/sh" "-c" "governance slack-join --channel C041U79AZAT"]
null_resource.slack_bridge_login_org_outreach: Creation complete after 0s [id=3487305657818532648]
synapse_mautrix_slack_link.org_outreach: Creating...
null_resource.slack_bridge_login_org_labrador: Creation complete after 0s [id=4972031203144043103]
synapse_mautrix_slack_link.org_labrador: Creating...
synapse_mautrix_slack_link.org_outreach: Still creating... [10s elapsed]
synapse_mautrix_slack_link.org_labrador: Still creating... [10s elapsed]
synapse_mautrix_slack_link.org_outreach: Creation complete after 20s [id=1549280019686101182]
synapse_mautrix_slack_link.org_labrador: Creation complete after 20s [id=1549280062581252187]

Apply complete! Resources: 4 added, 0 changed, 0 destroyed.

5. project: git dir: tofu/git workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.07s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
github_repository.slai_cmugpt_surface: Modifying... [id=cmugpt-surface]
github_repository.slai_cmugpt_agent: Modifying... [id=cmugpt-agent]
forgejo_team.components_leads: Creating...
forgejo_team.scottylabs-org_leads: Creating...
forgejo_team.lost-and-found-v2_leads: Creating...
github_team_membership.cbp_nikhilr: Creating...
github_team_membership.cmu-cal_anish: Creating...
forgejo_team_member.tartan-vote_xboxbedrock: Creating...
forgejo_team_member.cal_leads_anish: Creating...
forgejo_team_member.cbp_nikhilr: Creating...
forgejo_team_member.cal_leads_anish: Creation complete after 0s
forgejo_team_member.tartan-vote_xboxbedrock: Creation complete after 0s
forgejo_team_member.bus-sign_leads_anish: Creating...
forgejo_team_member.dalmatian_leads_anish: Creating...
forgejo_team_member.cbp_nikhilr: Creation complete after 0s
forgejo_team_member.slai_anish: Creating...
forgejo_team_member.bus-sign_leads_anish: Creation complete after 0s
forgejo_team_member.groupme-mirror_leads_anish: Creating...
forgejo_team.scottylabs-org_leads: Creation complete after 0s [name=scottylabs-org-leads]
forgejo_team.components_leads: Creation complete after 0s [name=components-leads]
forgejo_team_member.cmu-cal_anish: Creating...
forgejo_team_member.cmugpt-sms-surface_leads_anish: Creating...
forgejo_team.lost-and-found-v2_leads: Creation complete after 1s [name=lost-and-found-v2-leads]
forgejo_team_member.dalmatian_leads_anish: Creation complete after 1s
forgejo_team_member.tartan-vote_leads_anish: Creating...
forgejo_team_member.slai_anish: Creation complete after 1s
forgejo_team_member.lost-and-found_anish: Creating...
github_team_membership.tartan-vote_anish: Creating...
forgejo_team_member.groupme-mirror_leads_anish: Creation complete after 1s
forgejo_team_member.cmu-cal_anish: Creation complete after 1s
github_team_membership.lost-and-found_anish: Creating...
forgejo_team_member.slai_xboxbedrock: Creating...
forgejo_team_member.cmugpt-sms-surface_leads_anish: Creation complete after 1s
forgejo_team_member.tartan-vote_leads_anish: Creation complete after 0s
forgejo_team_member.lost-and-found_anish: Creation complete after 0s
github_team_membership.slai_anish: Creating...
github_team_membership.tartan-vote_tentype: Creating...
forgejo_team_member.slai_xboxbedrock: Creation complete after 0s
forgejo_team_member.tartan-vote_tentype: Creating...
github_team_membership.cmu-housing_anish: Creating...
forgejo_team_member.tartan-vote_tentype: Creation complete after 0s
forgejo_team_member.housing_leads_anish: Creating...
forgejo_team_member.housing_leads_anish: Creation complete after 0s
forgejo_team_member.discord-verify_leads_anish: Creating...
forgejo_team_member.discord-verify_leads_anish: Creation complete after 0s
forgejo_team_member.cbp_anish: Creating...
forgejo_team_member.cbp_anish: Creation complete after 0s
forgejo_team_member.cmugpt-mcp-server_leads_anish: Creating...
forgejo_team_member.cmugpt-mcp-server_leads_anish: Creation complete after 0s
forgejo_team_member.cmugpt-agent_leads_anish: Creating...
forgejo_team_member.cmugpt-agent_leads_anish: Creation complete after 0s
github_team_membership.uia_anish: Creating...
github_repository.slai_cmugpt_surface: Modifications complete after 7s [id=cmugpt-surface]
github_team_membership.slai_xboxbedrock: Creating...
github_repository.slai_cmugpt_agent: Still modifying... [id=cmugpt-agent, 10s elapsed]
github_team_membership.cbp_nikhilr: Still creating... [10s elapsed]
github_team_membership.cmu-cal_anish: Still creating... [10s elapsed]
github_team_membership.tartan-vote_anish: Still creating... [10s elapsed]
github_team_membership.lost-and-found_anish: Still creating... [10s elapsed]
github_team_membership.slai_anish: Still creating... [10s elapsed]
github_team_membership.tartan-vote_tentype: Still creating... [10s elapsed]
github_team_membership.cmu-housing_anish: Still creating... [10s elapsed]
github_team_membership.uia_anish: Still creating... [10s elapsed]
github_repository.slai_cmugpt_agent: Modifications complete after 14s [id=cmugpt-agent]
github_team_membership.cbp_xboxbedrock: Creating...
github_team_membership.cbp_nikhilr: Creation complete after 14s [id=17170099:ecstaticpilot]
forgejo_team_member.uia_anish: Creating...
forgejo_team_member.uia_anish: Creation complete after 0s
github_team_membership.tartan-vote_xboxbedrock: Creating...
github_team_membership.cmu-cal_anish: Creation complete after 14s [id=15669112:ap-1]
forgejo_team_member.cmugpt-surface_leads_anish: Creating...
forgejo_team_member.cmugpt-surface_leads_anish: Creation complete after 0s
forgejo_team_member.cbp_xboxbedrock: Creating...
forgejo_team_member.cbp_xboxbedrock: Creation complete after 0s
forgejo_team_member.cmu-housing_anish: Creating...
forgejo_team_member.cmu-housing_anish: Creation complete after 0s
forgejo_team_member.tartan-vote_anish: Creating...
forgejo_team_member.tartan-vote_anish: Creation complete after 0s
github_team_membership.cbp_anish: Creating...
github_team_membership.tartan-vote_anish: Creation complete after 13s [id=17701833:ap-1]
forgejo_team_member.scottylabs-org_leads_anish: Creating...
forgejo_team_member.scottylabs-org_leads_anish: Creation complete after 0s
forgejo_team_repository.scottylabs-org_leads_scottylabs_org: Creating...
forgejo_team_repository.scottylabs-org_leads_scottylabs_org: Creation complete after 0s
forgejo_team_repository.components_leads_components: Creating...
forgejo_team_repository.components_leads_components: Creation complete after 0s
forgejo_team_member.components_leads_anish: Creating...
forgejo_team_member.components_leads_anish: Creation complete after 0s
forgejo_team_repository.lost-and-found-v2_leads_lost_and_found_v2: Creating...
forgejo_team_repository.lost-and-found-v2_leads_lost_and_found_v2: Creation complete after 1s
forgejo_team_member.lost-and-found-v2_leads_anish: Creating...
forgejo_team_member.lost-and-found-v2_leads_anish: Creation complete after 0s
github_team_membership.lost-and-found_anish: Creation complete after 15s [id=4088608:ap-1]
github_team_membership.slai_anish: Creation complete after 15s [id=17571328:ap-1]
github_team_membership.tartan-vote_tentype: Creation complete after 15s [id=17701833:tentype]
github_team_membership.cmu-housing_anish: Creation complete after 15s [id=17937919:ap-1]
github_team_membership.uia_anish: Creation complete after 15s [id=17529763:ap-1]
github_team_membership.slai_xboxbedrock: Still creating... [10s elapsed]
github_team_membership.slai_xboxbedrock: Creation complete after 14s [id=17571328:xboxbedrock]
github_team_membership.cbp_xboxbedrock: Creation complete after 7s [id=17170099:xboxbedrock]
github_team_membership.tartan-vote_xboxbedrock: Creation complete after 7s [id=17701833:xboxbedrock]
github_team_membership.cbp_anish: Creation complete after 7s [id=17170099:ap-1]

Apply complete! Resources: 44 added, 2 changed, 0 destroyed.

Apply Summary

5 projects, 5 successful, 0 failed, 0 errored

Ran Apply for 5 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.19s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_cbp_ch0_nikhilr: Creating... null_resource.slack_cbp_ch0_anish: Creating... null_resource.slack_lost-and-found_ch0_anish: Creating... null_resource.slack_lost-and-found_ch0_anish: Provisioning with 'local-exec'... null_resource.slack_lost-and-found_ch0_anish (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C01BAADS8U8 --user U07FPJKFB5E"] null_resource.slack_slai_ch0_anish: Creating... null_resource.slack_cbp_ch0_xboxbedrock: Creating... null_resource.slack_slai_ch0_xboxbedrock: Creating... null_resource.slack_cbp_ch0_nikhilr: Provisioning with 'local-exec'... null_resource.slack_cbp_ch0_nikhilr (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08S56CJ8R2 --user U0AMBE2QKBM"] null_resource.slack_cbp_ch0_anish: Provisioning with 'local-exec'... null_resource.slack_cbp_ch0_anish (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08S56CJ8R2 --user U07FPJKFB5E"] null_resource.slack_cbp_ch0_xboxbedrock: Provisioning with 'local-exec'... null_resource.slack_cbp_ch0_xboxbedrock (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08S56CJ8R2 --user U0AQRKYLGHK"] null_resource.slack_uia_ch0_anish: Creating... null_resource.slack_cmu-cal_ch0_anish: Creating... null_resource.slack_slai_ch0_anish: Provisioning with 'local-exec'... null_resource.slack_slai_ch0_anish (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08SRUGUQ05 --user U07FPJKFB5E"] null_resource.slack_slai_ch0_xboxbedrock: Provisioning with 'local-exec'... null_resource.slack_slai_ch0_xboxbedrock (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08SRUGUQ05 --user U0AQRKYLGHK"] null_resource.slack_uia_ch0_anish: Provisioning with 'local-exec'... null_resource.slack_uia_ch0_anish (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C08MAT652CE --user U07FPJKFB5E"] null_resource.slack_cmu-cal_ch0_anish: Provisioning with 'local-exec'... null_resource.slack_cmu-cal_ch0_anish (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C05TNBPHVTL --user U07FPJKFB5E"] null_resource.slack_slai_ch0_anish: Creation complete after 0s [id=5303212720340085892] null_resource.slack_slai_ch0_xboxbedrock: Creation complete after 0s [id=7816746119839820188] null_resource.slack_uia_ch0_anish: Creation complete after 0s [id=4805975139938897672] null_resource.slack_cmu-cal_ch0_anish: Creation complete after 0s [id=3978892587047122247] null_resource.slack_cbp_ch0_anish: Creation complete after 0s [id=2983489810514487445] null_resource.slack_cbp_ch0_nikhilr: Creation complete after 0s [id=4626205003162193013] null_resource.slack_lost-and-found_ch0_anish: Creation complete after 0s [id=5751452929757186483] null_resource.slack_cbp_ch0_xboxbedrock: Creation complete after 1s [id=1848007231579617513] Apply complete! Resources: 8 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.20s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.nikhilr: Modifying... [id=1353118209024331948:795153900553764874] discord_member_roles.tentype: Modifying... [id=1353118209024331948:637528457508552705] discord_member_roles.anish: Modifying... [id=1353118209024331948:315249559473225729] discord_member_roles.xboxbedrock: Modifying... [id=1353118209024331948:712398210345205871] discord_member_roles.nikhilr: Modifications complete after 1s [id=1353118209024331948:795153900553764874] discord_member_roles.tentype: Modifications complete after 1s [id=1353118209024331948:637528457508552705] discord_member_roles.anish: Modifications complete after 2s [id=1353118209024331948:315249559473225729] discord_member_roles.xboxbedrock: Modifications complete after 2s [id=1353118209024331948:712398210345205871] Apply complete! Resources: 0 added, 4 changed, 0 destroyed. ``` </details> --- ### 3. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.21s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cbp_dalmatian_admins: Modifying... [id=scottylabs/group-memberships/5467214e-843c-488a-85fb-2ca5fa77d7d7] keycloak_group_memberships.project_cbp_dalmatian: Modifying... [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153] keycloak_group_memberships.project_uia_components: Modifying... [id=scottylabs/group-memberships/95465a93-b14a-459d-85b5-12f8b0fe7fd2] keycloak_group_memberships.project_slai_cmugpt-sms-surface: Modifying... [id=scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43] keycloak_group_memberships.project_lost-and-found_lost-and-found-v2_admins: Modifying... [id=scottylabs/group-memberships/7879878f-16a8-4b07-835e-81a8a9e7b02f] keycloak_group_memberships.project_tartan-vote_admins: Modifying... [id=scottylabs/group-memberships/cee98bbd-fb26-4527-ac5d-8a8587de2416] keycloak_group_memberships.project_slai_cmugpt-agent: Modifying... [id=scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423] keycloak_group_memberships.project_uia_components_admins: Modifying... [id=scottylabs/group-memberships/75073d65-02c5-405b-bc54-b4036dbda934] keycloak_group_memberships.project_slai_cmugpt-mcp-server_admins: Modifying... [id=scottylabs/group-memberships/8f84f9ec-e78a-41f6-9ce6-097fce273298] keycloak_group_memberships.project_slai_cmugpt-agent_admins: Modifying... [id=scottylabs/group-memberships/f4e82f9b-769d-4f38-a7f2-e214189d9269] keycloak_group_memberships.project_slai_cmugpt-agent_admins: Modifications complete after 2s [id=scottylabs/group-memberships/f4e82f9b-769d-4f38-a7f2-e214189d9269] keycloak_group_memberships.project_cmu-cal_cal: Modifying... [id=scottylabs/group-memberships/3800a836-0a57-40b7-839f-650a335bfccc] keycloak_group_memberships.project_cbp_dalmatian_admins: Modifications complete after 2s [id=scottylabs/group-memberships/5467214e-843c-488a-85fb-2ca5fa77d7d7] keycloak_group_memberships.project_slai_cmugpt-mcp-server: Modifying... [id=scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8] keycloak_group_memberships.project_uia_components_admins: Modifications complete after 2s [id=scottylabs/group-memberships/75073d65-02c5-405b-bc54-b4036dbda934] keycloak_group_memberships.project_tartan-vote: Modifying... [id=scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c] keycloak_group_memberships.project_slai_cmugpt-mcp-server_admins: Modifications complete after 2s [id=scottylabs/group-memberships/8f84f9ec-e78a-41f6-9ce6-097fce273298] keycloak_group_memberships.project_slai_cmugpt-sms-surface_admins: Modifying... [id=scottylabs/group-memberships/34404457-985f-4a61-b2f8-4754c2eb3293] keycloak_group_memberships.project_uia_components: Modifications complete after 2s [id=scottylabs/group-memberships/95465a93-b14a-459d-85b5-12f8b0fe7fd2] keycloak_group_memberships.project_cbp_discord-verify: Modifying... [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745] keycloak_group_memberships.project_lost-and-found_lost-and-found-v2_admins: Modifications complete after 2s [id=scottylabs/group-memberships/7879878f-16a8-4b07-835e-81a8a9e7b02f] keycloak_group_memberships.project_cmu-cal_cal_admins: Modifying... [id=scottylabs/group-memberships/5ca15cf1-036f-4644-b45e-8ca25ce9b2af] keycloak_group_memberships.project_tartan-vote_admins: Modifications complete after 2s [id=scottylabs/group-memberships/cee98bbd-fb26-4527-ac5d-8a8587de2416] keycloak_group_memberships.project_cbp_discord-verify_admins: Modifying... [id=scottylabs/group-memberships/a2abc505-3248-42f8-9513-d3db72be5d40] keycloak_group_memberships.project_slai_cmugpt-agent: Modifications complete after 4s [id=scottylabs/group-memberships/9aad8c08-a972-439d-b29d-f3a7f5181423] keycloak_group_memberships.project_lost-and-found_lost-and-found-v2: Modifying... [id=scottylabs/group-memberships/c14b4061-e3db-42c9-9b9e-a0c0a9747ce4] keycloak_group_memberships.project_slai_cmugpt-sms-surface: Modifications complete after 4s [id=scottylabs/group-memberships/45c6df66-57c5-427f-8748-865054d41f43] keycloak_group_memberships.project_uia_scottylabs-org: Modifying... [id=scottylabs/group-memberships/f7ea42f9-6007-4a5d-b46b-e5ac1d3e2bb8] keycloak_group_memberships.project_cmu-cal_cal: Modifications complete after 2s [id=scottylabs/group-memberships/3800a836-0a57-40b7-839f-650a335bfccc] keycloak_group_memberships.project_uia_scottylabs-org_admins: Modifying... [id=scottylabs/group-memberships/5786d2ba-5816-4ce2-8f04-d76958c4ee91] keycloak_group_memberships.project_cmu-cal_cal_admins: Modifications complete after 2s [id=scottylabs/group-memberships/5ca15cf1-036f-4644-b45e-8ca25ce9b2af] keycloak_group_memberships.project_cmu-housing_housing: Modifying... [id=scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d] keycloak_group_memberships.project_slai_cmugpt-sms-surface_admins: Modifications complete after 2s [id=scottylabs/group-memberships/34404457-985f-4a61-b2f8-4754c2eb3293] keycloak_group_memberships.project_cbp_groupme-mirror: Modifying... [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39] keycloak_group_memberships.project_cbp_discord-verify_admins: Modifications complete after 2s [id=scottylabs/group-memberships/a2abc505-3248-42f8-9513-d3db72be5d40] keycloak_group_memberships.project_slai_cmugpt-surface: Modifying... [id=scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775] keycloak_group_memberships.project_slai_cmugpt-mcp-server: Modifications complete after 4s [id=scottylabs/group-memberships/d833990f-eabb-4bfc-942f-4f7a7c48d0f8] keycloak_group_memberships.project_cbp_bus-sign: Modifying... [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b] keycloak_group_memberships.project_uia_scottylabs-org: Modifications complete after 2s [id=scottylabs/group-memberships/f7ea42f9-6007-4a5d-b46b-e5ac1d3e2bb8] keycloak_group_memberships.project_cmu-housing_housing_admins: Modifying... [id=scottylabs/group-memberships/7c932acc-ac1e-420b-8040-9c3af456b597] keycloak_group_memberships.project_lost-and-found_lost-and-found-v2: Modifications complete after 2s [id=scottylabs/group-memberships/c14b4061-e3db-42c9-9b9e-a0c0a9747ce4] keycloak_group_memberships.project_cbp_groupme-mirror_admins: Modifying... [id=scottylabs/group-memberships/c4c2efc2-d322-4193-8279-699e397522b0] keycloak_group_memberships.project_uia_scottylabs-org_admins: Modifications complete after 3s [id=scottylabs/group-memberships/5786d2ba-5816-4ce2-8f04-d76958c4ee91] keycloak_group_memberships.project_slai_cmugpt-surface_admins: Modifying... [id=scottylabs/group-memberships/4f3531db-8075-480e-bbf4-253e98783b56] keycloak_group_memberships.project_cbp_dalmatian: Modifications complete after 7s [id=scottylabs/group-memberships/324ba2c4-7538-4050-98d8-9ffb07bb1153] keycloak_group_memberships.project_cbp_bus-sign_admins: Modifying... [id=scottylabs/group-memberships/100e7090-c4a4-47f2-b0ad-8eb81dd8c8c8] keycloak_group_memberships.project_cmu-housing_housing: Modifications complete after 3s [id=scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d] keycloak_group_memberships.project_tartan-vote: Modifications complete after 6s [id=scottylabs/group-memberships/c32fb4e6-6198-4962-b0e5-a497a23dad8c] keycloak_group_memberships.project_cmu-housing_housing_admins: Modifications complete after 2s [id=scottylabs/group-memberships/7c932acc-ac1e-420b-8040-9c3af456b597] keycloak_group_memberships.project_slai_cmugpt-surface: Modifications complete after 4s [id=scottylabs/group-memberships/49b7dd85-aa5c-48bd-93c1-3fe377683775] keycloak_group_memberships.project_slai_cmugpt-surface_admins: Modifications complete after 2s [id=scottylabs/group-memberships/4f3531db-8075-480e-bbf4-253e98783b56] keycloak_group_memberships.project_cbp_groupme-mirror_admins: Modifications complete after 3s [id=scottylabs/group-memberships/c4c2efc2-d322-4193-8279-699e397522b0] keycloak_group_memberships.project_cbp_discord-verify: Modifications complete after 7s [id=scottylabs/group-memberships/295be9ba-bd62-4f8c-81a0-ee49278f3745] keycloak_group_memberships.project_cbp_bus-sign_admins: Modifications complete after 2s [id=scottylabs/group-memberships/100e7090-c4a4-47f2-b0ad-8eb81dd8c8c8] keycloak_group_memberships.project_cbp_groupme-mirror: Modifications complete after 6s [id=scottylabs/group-memberships/e99499c1-c5e2-4121-a207-ea138dce2e39] keycloak_group_memberships.project_cbp_bus-sign: Modifications complete after 6s [id=scottylabs/group-memberships/7dd87f8f-e735-4316-af11-560d6ef9292b] Apply complete! Resources: 0 added, 28 changed, 0 destroyed. ``` </details> --- ### 4. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.12s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_bridge_login_org_labrador: Creating... null_resource.slack_bridge_login_org_labrador: Provisioning with 'local-exec'... null_resource.slack_bridge_login_org_labrador (local-exec): Executing: ["/bin/sh" "-c" "governance slack-join --channel C05TG63EM9B"] null_resource.slack_bridge_login_org_outreach: Creating... null_resource.slack_bridge_login_org_outreach: Provisioning with 'local-exec'... null_resource.slack_bridge_login_org_outreach (local-exec): Executing: ["/bin/sh" "-c" "governance slack-join --channel C041U79AZAT"] null_resource.slack_bridge_login_org_outreach: Creation complete after 0s [id=3487305657818532648] synapse_mautrix_slack_link.org_outreach: Creating... null_resource.slack_bridge_login_org_labrador: Creation complete after 0s [id=4972031203144043103] synapse_mautrix_slack_link.org_labrador: Creating... synapse_mautrix_slack_link.org_outreach: Still creating... [10s elapsed] synapse_mautrix_slack_link.org_labrador: Still creating... [10s elapsed] synapse_mautrix_slack_link.org_outreach: Creation complete after 20s [id=1549280019686101182] synapse_mautrix_slack_link.org_labrador: Creation complete after 20s [id=1549280062581252187] Apply complete! Resources: 4 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.07s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ github_repository.slai_cmugpt_surface: Modifying... [id=cmugpt-surface] github_repository.slai_cmugpt_agent: Modifying... [id=cmugpt-agent] forgejo_team.components_leads: Creating... forgejo_team.scottylabs-org_leads: Creating... forgejo_team.lost-and-found-v2_leads: Creating... github_team_membership.cbp_nikhilr: Creating... github_team_membership.cmu-cal_anish: Creating... forgejo_team_member.tartan-vote_xboxbedrock: Creating... forgejo_team_member.cal_leads_anish: Creating... forgejo_team_member.cbp_nikhilr: Creating... forgejo_team_member.cal_leads_anish: Creation complete after 0s forgejo_team_member.tartan-vote_xboxbedrock: Creation complete after 0s forgejo_team_member.bus-sign_leads_anish: Creating... forgejo_team_member.dalmatian_leads_anish: Creating... forgejo_team_member.cbp_nikhilr: Creation complete after 0s forgejo_team_member.slai_anish: Creating... forgejo_team_member.bus-sign_leads_anish: Creation complete after 0s forgejo_team_member.groupme-mirror_leads_anish: Creating... forgejo_team.scottylabs-org_leads: Creation complete after 0s [name=scottylabs-org-leads] forgejo_team.components_leads: Creation complete after 0s [name=components-leads] forgejo_team_member.cmu-cal_anish: Creating... forgejo_team_member.cmugpt-sms-surface_leads_anish: Creating... forgejo_team.lost-and-found-v2_leads: Creation complete after 1s [name=lost-and-found-v2-leads] forgejo_team_member.dalmatian_leads_anish: Creation complete after 1s forgejo_team_member.tartan-vote_leads_anish: Creating... forgejo_team_member.slai_anish: Creation complete after 1s forgejo_team_member.lost-and-found_anish: Creating... github_team_membership.tartan-vote_anish: Creating... forgejo_team_member.groupme-mirror_leads_anish: Creation complete after 1s forgejo_team_member.cmu-cal_anish: Creation complete after 1s github_team_membership.lost-and-found_anish: Creating... forgejo_team_member.slai_xboxbedrock: Creating... forgejo_team_member.cmugpt-sms-surface_leads_anish: Creation complete after 1s forgejo_team_member.tartan-vote_leads_anish: Creation complete after 0s forgejo_team_member.lost-and-found_anish: Creation complete after 0s github_team_membership.slai_anish: Creating... github_team_membership.tartan-vote_tentype: Creating... forgejo_team_member.slai_xboxbedrock: Creation complete after 0s forgejo_team_member.tartan-vote_tentype: Creating... github_team_membership.cmu-housing_anish: Creating... forgejo_team_member.tartan-vote_tentype: Creation complete after 0s forgejo_team_member.housing_leads_anish: Creating... forgejo_team_member.housing_leads_anish: Creation complete after 0s forgejo_team_member.discord-verify_leads_anish: Creating... forgejo_team_member.discord-verify_leads_anish: Creation complete after 0s forgejo_team_member.cbp_anish: Creating... forgejo_team_member.cbp_anish: Creation complete after 0s forgejo_team_member.cmugpt-mcp-server_leads_anish: Creating... forgejo_team_member.cmugpt-mcp-server_leads_anish: Creation complete after 0s forgejo_team_member.cmugpt-agent_leads_anish: Creating... forgejo_team_member.cmugpt-agent_leads_anish: Creation complete after 0s github_team_membership.uia_anish: Creating... github_repository.slai_cmugpt_surface: Modifications complete after 7s [id=cmugpt-surface] github_team_membership.slai_xboxbedrock: Creating... github_repository.slai_cmugpt_agent: Still modifying... [id=cmugpt-agent, 10s elapsed] github_team_membership.cbp_nikhilr: Still creating... [10s elapsed] github_team_membership.cmu-cal_anish: Still creating... [10s elapsed] github_team_membership.tartan-vote_anish: Still creating... [10s elapsed] github_team_membership.lost-and-found_anish: Still creating... [10s elapsed] github_team_membership.slai_anish: Still creating... [10s elapsed] github_team_membership.tartan-vote_tentype: Still creating... [10s elapsed] github_team_membership.cmu-housing_anish: Still creating... [10s elapsed] github_team_membership.uia_anish: Still creating... [10s elapsed] github_repository.slai_cmugpt_agent: Modifications complete after 14s [id=cmugpt-agent] github_team_membership.cbp_xboxbedrock: Creating... github_team_membership.cbp_nikhilr: Creation complete after 14s [id=17170099:ecstaticpilot] forgejo_team_member.uia_anish: Creating... forgejo_team_member.uia_anish: Creation complete after 0s github_team_membership.tartan-vote_xboxbedrock: Creating... github_team_membership.cmu-cal_anish: Creation complete after 14s [id=15669112:ap-1] forgejo_team_member.cmugpt-surface_leads_anish: Creating... forgejo_team_member.cmugpt-surface_leads_anish: Creation complete after 0s forgejo_team_member.cbp_xboxbedrock: Creating... forgejo_team_member.cbp_xboxbedrock: Creation complete after 0s forgejo_team_member.cmu-housing_anish: Creating... forgejo_team_member.cmu-housing_anish: Creation complete after 0s forgejo_team_member.tartan-vote_anish: Creating... forgejo_team_member.tartan-vote_anish: Creation complete after 0s github_team_membership.cbp_anish: Creating... github_team_membership.tartan-vote_anish: Creation complete after 13s [id=17701833:ap-1] forgejo_team_member.scottylabs-org_leads_anish: Creating... forgejo_team_member.scottylabs-org_leads_anish: Creation complete after 0s forgejo_team_repository.scottylabs-org_leads_scottylabs_org: Creating... forgejo_team_repository.scottylabs-org_leads_scottylabs_org: Creation complete after 0s forgejo_team_repository.components_leads_components: Creating... forgejo_team_repository.components_leads_components: Creation complete after 0s forgejo_team_member.components_leads_anish: Creating... forgejo_team_member.components_leads_anish: Creation complete after 0s forgejo_team_repository.lost-and-found-v2_leads_lost_and_found_v2: Creating... forgejo_team_repository.lost-and-found-v2_leads_lost_and_found_v2: Creation complete after 1s forgejo_team_member.lost-and-found-v2_leads_anish: Creating... forgejo_team_member.lost-and-found-v2_leads_anish: Creation complete after 0s github_team_membership.lost-and-found_anish: Creation complete after 15s [id=4088608:ap-1] github_team_membership.slai_anish: Creation complete after 15s [id=17571328:ap-1] github_team_membership.tartan-vote_tentype: Creation complete after 15s [id=17701833:tentype] github_team_membership.cmu-housing_anish: Creation complete after 15s [id=17937919:ap-1] github_team_membership.uia_anish: Creation complete after 15s [id=17529763:ap-1] github_team_membership.slai_xboxbedrock: Still creating... [10s elapsed] github_team_membership.slai_xboxbedrock: Creation complete after 14s [id=17571328:xboxbedrock] github_team_membership.cbp_xboxbedrock: Creation complete after 7s [id=17170099:xboxbedrock] github_team_membership.tartan-vote_xboxbedrock: Creation complete after 7s [id=17701833:xboxbedrock] github_team_membership.cbp_anish: Creation complete after 7s [id=17170099:ap-1] Apply complete! Resources: 44 added, 2 changed, 0 destroyed. ``` </details> --- ### Apply Summary 5 projects, 5 successful, 0 failed, 0 errored
anish merged commit 3d5c9bbce6 into main 2026-09-15 05:11:51 +00:00
anish deleted branch committees 2026-09-15 05:11:52 +00:00

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/matrix_bridges workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/matrix_bridges` workspace: `default` - dir: `tofu/slack` workspace: `default`
Sign in to join this conversation.
No description provided.