feat: add cmu-eats team members #192

Open
arom1a wants to merge 4 commits from arom1a/governance:main into main
First-time contributor
No description provided.
feat: add cmu-eats team members
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Failing after 4m27s
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
CI / build (pull_request) Has been cancelled
d844568f53
arom1a force-pushed main from d844568f53
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Failing after 4m27s
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
CI / build (pull_request) Has been cancelled
to a969de2431
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Failing after 4m30s
CI / check (pull_request) Has been cancelled
CI / check-1 (pull_request) Has been cancelled
CI / build (pull_request) Has been cancelled
2026-09-12 20:01:39 +00:00
Compare
chore: remove unregistered cmu-eats members
Some checks failed
atlantis/plan Plan failed.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Failing after 4m22s
CI / build (pull_request) Has been skipped
CI / check-1 (pull_request) Failing after 5m27s
CI / check (pull_request) Failing after 0s
196a1228cd
Error: Pre-workflow hook failed: running git clone --depth=1 --branch main --single-branch https://scottylabs-bot:<redacted>@git.cmu.dev/arom1a/governance.git /var/lib/atlantis/repos/ScottyLabs/governance/192/default: Cloning into '/var/lib/atlantis/repos/ScottyLabs/governance/192/default'...
remote: Repository not found
fatal: repository 'https://git.cmu.dev/arom1a/governance.git/' not found
: exit status 128
``` Error: Pre-workflow hook failed: running git clone --depth=1 --branch main --single-branch https://scottylabs-bot:<redacted>@git.cmu.dev/arom1a/governance.git /var/lib/atlantis/repos/ScottyLabs/governance/192/default: Cloning into '/var/lib/atlantis/repos/ScottyLabs/governance/192/default'... remote: Repository not found fatal: repository 'https://git.cmu.dev/arom1a/governance.git/' not found : exit status 128 ```
Owner
error: keycloak API error: forgejo user lookup for arom1a: http status: 404
error: user cirex-web has not linked their github account in keycloak
error: user cirex-web has not linked their slack account in keycloak

see step 1 from readme

``` error: keycloak API error: forgejo user lookup for arom1a: http status: 404 error: user cirex-web has not linked their github account in keycloak error: user cirex-web has not linked their slack account in keycloak ``` see step 1 from readme
Author
First-time contributor

fixed

fixed
arom1a force-pushed main from 196a1228cd
Some checks failed
atlantis/plan Plan failed.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Failing after 4m22s
CI / build (pull_request) Has been skipped
CI / check-1 (pull_request) Failing after 5m27s
CI / check (pull_request) Failing after 0s
to d6d21bf0ac
Some checks failed
CI / build (pull_request) Has been skipped
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
Validate / Governance (pull_request_target) Failing after 4m36s
CI / check-1 (pull_request) Failing after 5m54s
CI / check (pull_request) Failing after 0s
2026-09-16 18:18:14 +00:00
Compare
Author
First-time contributor

image

idk, and no idea abt the first error

![image](/attachments/32be23b3-decf-4d54-9b37-dd11a422c878) idk, and no idea abt the first error
Owner
error: keycloak API error: forgejo user lookup for arom1a: http status: 404
error: user cirex-web has not linked their github account in keycloak

image

it seems like eric is missing github and google, and you are missing google

the first msg is probably a transient error

``` error: keycloak API error: forgejo user lookup for arom1a: http status: 404 error: user cirex-web has not linked their github account in keycloak ``` ![image](/attachments/f8e88e1d-863a-4810-8d6a-ac8acd4fd34b) it seems like eric is missing github and google, and you are missing google the first msg is probably a transient error
chore: remove cirex
Some checks failed
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Failing after 4m22s
CI / check-1 (pull_request) Failing after 5m51s
CI / check (pull_request) Failing after 0s
94c509fd7e
anish force-pushed main from 94c509fd7e
Some checks failed
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Failing after 4m22s
CI / check-1 (pull_request) Failing after 5m51s
CI / check (pull_request) Failing after 0s
to d1576de381
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
CI / build (pull_request) Has been skipped
kennel/build skipped: no services or static sites defined
CI / check-1 (pull_request) Failing after 12m42s
CI / check (pull_request) Failing after 0s
Validate / Governance (pull_request_target) Successful in 2m55s
2026-09-18 18:36:08 +00:00
Compare
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: keycloak Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan 6/6 projects failed to plan.
kennel/build skipped: no services or static sites defined
64d76fa731

Ran Plan for 6 projects:

  1. project: keycloak dir: tofu/keycloak workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: slack dir: tofu/slack workspace: default
  4. project: git dir: tofu/git workspace: default
  5. project: google_play dir: tofu/google_play workspace: default
  6. project: google_groups dir: tofu/google_groups workspace: default

1. project: keycloak dir: tofu/keycloak workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


2. project: discord dir: tofu/discord workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


3. project: slack dir: tofu/slack workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


4. project: git dir: tofu/git workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


5. project: google_play dir: tofu/google_play workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


6. project: google_groups dir: tofu/google_groups workspace: default

Plan Failed: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request.

Once the lock is released, comment atlantis plan here to re-plan.


Plan Summary

6 projects, 0 with changes, 0 with no changes, 6 failed

  • To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` --- ### 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 3. project: `slack` dir: `tofu/slack` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 4. project: `git` dir: `tofu/git` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 5. project: `google_play` dir: `tofu/google_play` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### 6. project: `google_groups` dir: `tofu/google_groups` workspace: `default` **Plan Failed**: This project is currently locked by an unapplied plan from pull #205. To continue, delete the lock from #205 or apply that plan and merge the pull request. Once the lock is released, comment `atlantis plan` here to re-plan. --- ### Plan Summary 6 projects, 0 with changes, 0 with no changes, 6 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan: keycloak Plan failed.
atlantis/plan: discord Plan failed.
atlantis/plan: slack Plan failed.
atlantis/plan: git Plan failed.
atlantis/plan: google_play Plan failed.
atlantis/plan: google_groups Plan failed.
atlantis/plan 6/6 projects failed to plan.
kennel/build skipped: no services or static sites defined
This pull request can be merged automatically.
This branch is out-of-date with the base branch
You are not authorized to merge this pull request.
View command line instructions

Checkout

From your project repository, check out a new branch and test the changes.
git fetch -u main:arom1a-main
git switch arom1a-main

Merge

Merge the changes and update on Forgejo.

Warning: The "Autodetect manual merge" setting is not enabled for this repository, you will have to mark this pull request as manually merged afterwards.

git switch main
git merge --no-ff arom1a-main
git switch arom1a-main
git rebase main
git switch main
git merge --ff-only arom1a-main
git switch arom1a-main
git rebase main
git switch main
git merge --no-ff arom1a-main
git switch main
git merge --squash arom1a-main
git switch main
git merge --ff-only arom1a-main
git switch main
git merge arom1a-main
git push origin main
Sign in to join this conversation.
No description provided.