Add Jessica to CMU Housing #181

Merged
anish merged 2 commits from jessica-lr-song/governance:main into main 2026-08-30 06:02:18 +00:00
No description provided.
Add Jessica
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Failing after 4m29s
CI / build (pull_request) Has been skipped
CI / check-1 (pull_request) Successful in 6m29s
CI / check (pull_request) Successful in 0s
f97bd2e374
anish left a comment

error: user jessica-lr-song not found in keycloak (no linked forgejo account)

error: user jessica-lr-song not found in keycloak (no linked forgejo account)
Author
Member

@anish wrote in #181 (comment):

error: user jessica-lr-song not found in keycloak (no linked forgejo account)

try again, should be good now

@anish wrote in https://git.cmu.dev/ScottyLabs/governance/pulls/181#issuecomment-2646: > error: user jessica-lr-song not found in keycloak (no linked forgejo account) try again, should be good now
anish force-pushed main from f97bd2e374
Some checks failed
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Failing after 4m29s
CI / build (pull_request) Has been skipped
CI / check-1 (pull_request) Successful in 6m29s
CI / check (pull_request) Successful in 0s
to 953dbb1308
All checks were successful
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 8m42s
CI / check-1 (pull_request) Successful in 12m31s
CI / check (pull_request) Successful in 0s
2026-08-30 05:53:56 +00:00
Compare
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
All checks were successful
atlantis/plan: slack Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: google_play No changes. Your infrastructure matches the configuration.
atlantis/plan: google_groups Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 0 to add, 2 to change, 0 to destroy.
atlantis/plan: discord Plan: 1 to add, 0 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
atlantis/plan: git Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan 6/6 projects planned successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: google_play Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: google_groups Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 6/6 projects applied successfully.
87cb172be0

Ran Plan for 6 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: google_groups dir: tofu/google_groups workspace: default
  4. project: keycloak dir: tofu/keycloak workspace: default
  5. project: discord dir: tofu/discord workspace: default
  6. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_hub_jessica_lr_song will be created
+ resource "null_resource" "slack_hub_jessica_lr_song" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C096TM8EMS8"
          + "user"    = "U0BATK9SG10"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 1 to add, 0 to change, 0 to destroy.


2. project: google_play dir: tofu/google_play workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.19s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

3. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.25s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # google_cloud_identity_group_membership.tech_jessica_lr_song will be created
+ resource "google_cloud_identity_group_membership" "tech_jessica_lr_song" {
+       create_time = (known after apply)
+       group       = "groups/04anzqyu1bqbv79"
+       id          = (known after apply)
+       name        = (known after apply)
+       type        = (known after apply)
+       update_time = (known after apply)

+       preferred_member_key {
+           id = "jsong5@andrew.cmu.edu"
        }

+       roles {
+           name = "MEMBER"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_groups
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_groups
    

Plan: 1 to add, 0 to change, 0 to destroy.


4. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.22s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place
! resource "keycloak_group_memberships" "project_cmu-housing_housing" {
        id       = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d"
!       members  = [
+           "jsong5",
            # (9 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

  # keycloak_group_memberships.projects will be updated in-place
! resource "keycloak_group_memberships" "projects" {
        id       = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25"
!       members  = [
+           "jsong5",
            # (27 unchanged elements hidden)
        ]
        # (2 unchanged attributes hidden)
    }

Plan: 0 to add, 2 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 0 to add, 2 to change, 0 to destroy.


5. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # discord_member_roles.jessica_lr_song will be created
+ resource "discord_member_roles" "jessica_lr_song" {
+       id        = (known after apply)
+       server_id = "1353118209024331948"
+       user_id   = "734975928366661714"

+       role {
+           has_role = true
+           role_id  = "1488787088118644848"
        }
+       role {
+           has_role = true
+           role_id  = "1513653499957874871"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 1 to add, 0 to change, 0 to destroy.


6. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.21s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # forgejo_team_member.cmu-housing_jessica_lr_song will be created
+ resource "forgejo_team_member" "cmu-housing_jessica_lr_song" {
+       team_id = 2
+       user    = "jessica-lr-song"
    }

  # github_team_membership.cmu-housing_jessica_lr_song will be created
+ resource "github_team_membership" "cmu-housing_jessica_lr_song" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "17937919"
+       username = "jessica-lr-song"
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 0 to change, 0 to destroy.


Plan Summary

6 projects, 5 with changes, 1 with no changes, 0 failed

  • To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 6 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_hub_jessica_lr_song will be created + resource "null_resource" "slack_hub_jessica_lr_song" { + id = (known after apply) + triggers = { + "channel" = "C096TM8EMS8" + "user" = "U0BATK9SG10" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.19s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` --- ### 3. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.25s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # google_cloud_identity_group_membership.tech_jessica_lr_song will be created + resource "google_cloud_identity_group_membership" "tech_jessica_lr_song" { + create_time = (known after apply) + group = "groups/04anzqyu1bqbv79" + id = (known after apply) + name = (known after apply) + type = (known after apply) + update_time = (known after apply) + preferred_member_key { + id = "jsong5@andrew.cmu.edu" } + roles { + name = "MEMBER" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_groups ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_groups%252Fdefault%252Fgoogle_groups) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_groups ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 4. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.22s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: ! update in-place (current -> planned) OpenTofu will perform the following actions: # keycloak_group_memberships.project_cmu-housing_housing will be updated in-place ! resource "keycloak_group_memberships" "project_cmu-housing_housing" { id = "scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d" ! members = [ + "jsong5", # (9 unchanged elements hidden) ] # (2 unchanged attributes hidden) } # keycloak_group_memberships.projects will be updated in-place ! resource "keycloak_group_memberships" "projects" { id = "scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25" ! members = [ + "jsong5", # (27 unchanged elements hidden) ] # (2 unchanged attributes hidden) } Plan: 0 to add, 2 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 0 to add, 2 to change, 0 to destroy. --- ### 5. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # discord_member_roles.jessica_lr_song will be created + resource "discord_member_roles" "jessica_lr_song" { + id = (known after apply) + server_id = "1353118209024331948" + user_id = "734975928366661714" + role { + has_role = true + role_id = "1488787088118644848" } + role { + has_role = true + role_id = "1513653499957874871" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 6. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.21s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # forgejo_team_member.cmu-housing_jessica_lr_song will be created + resource "forgejo_team_member" "cmu-housing_jessica_lr_song" { + team_id = 2 + user = "jessica-lr-song" } # github_team_membership.cmu-housing_jessica_lr_song will be created + resource "github_team_membership" "cmu-housing_jessica_lr_song" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "17937919" + username = "jessica-lr-song" } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### Plan Summary 6 projects, 5 with changes, 1 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Owner

atlantis apply

atlantis apply

Ran Apply for 6 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: discord dir: tofu/discord workspace: default
  4. project: google_groups dir: tofu/google_groups workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: keycloak dir: tofu/keycloak workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.10s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

2. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.20s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_hub_jessica_lr_song: Creating...
null_resource.slack_hub_jessica_lr_song: Provisioning with 'local-exec'...
null_resource.slack_hub_jessica_lr_song (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0BATK9SG10"]
null_resource.slack_hub_jessica_lr_song: Creation complete after 0s [id=5896781438471984357]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

3. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.16s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_member_roles.jessica_lr_song: Creating...
discord_member_roles.jessica_lr_song: Creation complete after 1s [id=1353118209024331948:734975928366661714]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

4. project: google_groups dir: tofu/google_groups workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.28s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
google_cloud_identity_group_membership.tech_jessica_lr_song: Creating...
google_cloud_identity_group_membership.tech_jessica_lr_song: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/110846605425290983348]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.26s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
forgejo_team_member.cmu-housing_jessica_lr_song: Creating...
forgejo_team_member.cmu-housing_jessica_lr_song: Creation complete after 0s
github_team_membership.cmu-housing_jessica_lr_song: Creating...
github_team_membership.cmu-housing_jessica_lr_song: Creation complete after 1s [id=17937919:jessica-lr-song]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

6. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.26s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group_memberships.project_cmu-housing_housing: Modifying... [id=scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d]
keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]
keycloak_group_memberships.project_cmu-housing_housing: Modifications complete after 3s [id=scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d]
keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25]

Apply complete! Resources: 0 added, 2 changed, 0 destroyed.

Apply Summary

6 projects, 6 successful, 0 failed, 0 errored

Ran Apply for 6 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `google_groups` dir: `tofu/google_groups` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.10s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.20s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_hub_jessica_lr_song: Creating... null_resource.slack_hub_jessica_lr_song: Provisioning with 'local-exec'... null_resource.slack_hub_jessica_lr_song (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C096TM8EMS8 --user U0BATK9SG10"] null_resource.slack_hub_jessica_lr_song: Creation complete after 0s [id=5896781438471984357] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.16s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_member_roles.jessica_lr_song: Creating... discord_member_roles.jessica_lr_song: Creation complete after 1s [id=1353118209024331948:734975928366661714] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `google_groups` dir: `tofu/google_groups` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.28s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ google_cloud_identity_group_membership.tech_jessica_lr_song: Creating... google_cloud_identity_group_membership.tech_jessica_lr_song: Creation complete after 1s [id=groups/04anzqyu1bqbv79/memberships/110846605425290983348] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.26s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ forgejo_team_member.cmu-housing_jessica_lr_song: Creating... forgejo_team_member.cmu-housing_jessica_lr_song: Creation complete after 0s github_team_membership.cmu-housing_jessica_lr_song: Creating... github_team_membership.cmu-housing_jessica_lr_song: Creation complete after 1s [id=17937919:jessica-lr-song] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 6. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.26s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group_memberships.project_cmu-housing_housing: Modifying... [id=scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d] keycloak_group_memberships.projects: Modifying... [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] keycloak_group_memberships.project_cmu-housing_housing: Modifications complete after 3s [id=scottylabs/group-memberships/a13c9974-065d-4dbc-b473-b238f545618d] keycloak_group_memberships.projects: Modifications complete after 3s [id=scottylabs/group-memberships/9e6890c8-50a0-4b82-86b9-91791fffac25] Apply complete! Resources: 0 added, 2 changed, 0 destroyed. ``` </details> --- ### Apply Summary 6 projects, 6 successful, 0 failed, 0 errored
anish merged commit 999250da9e into main 2026-08-30 06:02:18 +00:00
anish referenced this pull request from a commit 2026-08-30 06:02:19 +00:00

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_groups workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_groups` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/slack` workspace: `default`
Sign in to join this conversation.
No description provided.