feat: add google_play and cmu-eats, remove vaultwarden #178

Merged
anish merged 3 commits from google-play into main 2026-08-24 21:24:29 +00:00
Owner
No description provided.
feat: add google_play and cmu-eats, remove vaultwarden
Some checks failed
CI / build (pull_request) Has been skipped
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 4m43s
CI / check-1 (pull_request) Successful in 5m55s
CI / check (pull_request) Successful in 0s
atlantis/plan: google_play Plan: 11 to add, 0 to change, 0 to destroy.
atlantis/plan: slack Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/plan: openbao Plan: 6 to add, 0 to change, 0 to destroy.
atlantis/plan: posthog Plan: 3 to add, 0 to change, 0 to destroy.
atlantis/plan: matrix_bridges Plan: 2 to add, 0 to change, 0 to destroy.
atlantis/plan: sentry Plan: 4 to add, 0 to change, 0 to destroy.
atlantis/plan: discord Plan: 2 to add, 1 to change, 0 to destroy.
atlantis/plan: keycloak Plan: 38 to add, 0 to change, 0 to destroy.
atlantis/plan: git Plan: 16 to add, 27 to change, 0 to destroy.
atlantis/plan 9/9 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: google_play Apply failed.
atlantis/apply: git Apply failed.
atlantis/apply: slack Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: openbao Apply succeeded.
atlantis/apply: posthog Apply succeeded.
atlantis/apply: sentry Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply: matrix_bridges Apply succeeded.
atlantis/apply 2/9 projects failed to apply.
88ae3cd61a
Signed-off-by: Anish Pallati <i@anish.land>
Author
Owner

atlantis plan

atlantis plan
Author
Owner

atlantis plan

atlantis plan

Ran Plan for 9 projects:

  1. project: google_play dir: tofu/google_play workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: openbao dir: tofu/openbao workspace: default
  4. project: posthog dir: tofu/posthog workspace: default
  5. project: matrix_bridges dir: tofu/matrix_bridges workspace: default
  6. project: sentry dir: tofu/sentry workspace: default
  7. project: discord dir: tofu/discord workspace: default
  8. project: keycloak dir: tofu/keycloak workspace: default
  9. project: git dir: tofu/git workspace: default

1. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.35s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
data.external.identity_anish: Reading...
data.external.identity_thesuperrl: Reading...
data.external.identity_maybe_yiyi: Reading...
data.external.identity_liamwalk: Reading...
data.external.identity_tentype: Reading...
data.external.identity_brightz: Reading...
data.external.identity_jc_965: Reading...
data.external.identity_riztofu: Reading...
data.external.identity_kechezon: Reading...
data.external.identity_xboxbedrock: Reading...
data.external.identity_brightz: Read complete after 0s [id=-]
data.external.identity_kothurvidur: Reading...
data.external.identity_liamwalk: Read complete after 0s [id=-]
data.external.identity_jiyas: Reading...
data.external.identity_tentype: Read complete after 0s [id=-]
data.external.identity_dkou: Reading...
data.external.identity_jc_965: Read complete after 0s [id=-]
data.external.identity_rzhu3: Reading...
data.external.identity_riztofu: Read complete after 0s [id=-]
data.external.identity_maybe_yiyi: Read complete after 0s [id=-]
data.external.identity_krishsax: Reading...
data.external.identity_anish: Read complete after 0s [id=-]
data.external.identity_xboxbedrock: Read complete after 0s [id=-]
data.external.identity_thesuperrl: Read complete after 0s [id=-]
data.external.identity_kechezon: Read complete after 0s [id=-]
data.external.identity_jiyas: Read complete after 0s [id=-]
data.external.identity_kothurvidur: Read complete after 0s [id=-]
data.external.identity_rzhu3: Read complete after 0s [id=-]
data.external.identity_krishsax: Read complete after 0s [id=-]
data.external.identity_dkou: Read complete after 0s [id=-]

OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # googleplay_app_iam.quest_brightz will be created
+ resource "googleplay_app_iam" "quest_brightz" {
+       app_id               = "4972435805279727996"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "brightz@andrew.cmu.edu"
    }

  # googleplay_app_iam.quest_kechezon will be created
+ resource "googleplay_app_iam" "quest_kechezon" {
+       app_id               = "4972435805279727996"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "kechezon@andrew.cmu.edu"
    }

  # googleplay_app_iam.quest_xboxbedrock will be created
+ resource "googleplay_app_iam" "quest_xboxbedrock" {
+       app_id               = "4972435805279727996"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_APP_CONTENT",
+           "CAN_MANAGE_PUBLIC_APKS",
+           "CAN_MANAGE_PUBLIC_LISTING",
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_REPLY_TO_REVIEWS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "spola2@andrew.cmu.edu"
    }

  # googleplay_app_iam.terrier_dkou will be created
+ resource "googleplay_app_iam" "terrier_dkou" {
+       app_id               = "4975152612825535266"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "dkou@andrew.cmu.edu"
    }

  # googleplay_app_iam.terrier_kothurvidur will be created
+ resource "googleplay_app_iam" "terrier_kothurvidur" {
+       app_id               = "4975152612825535266"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "vkothur@andrew.cmu.edu"
    }

  # googleplay_user.anish will be created
+ resource "googleplay_user" "anish" {
+       email                = "apallati@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_MANAGE_PERMISSIONS_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

  # googleplay_user.brightz will be created
+ resource "googleplay_user" "brightz" {
+       email                = "brightz@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

  # googleplay_user.dkou will be created
+ resource "googleplay_user" "dkou" {
+       email                = "dkou@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

  # googleplay_user.kechezon will be created
+ resource "googleplay_user" "kechezon" {
+       email                = "kechezon@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

  # googleplay_user.kothurvidur will be created
+ resource "googleplay_user" "kothurvidur" {
+       email                = "vkothur@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

  # googleplay_user.xboxbedrock will be created
+ resource "googleplay_user" "xboxbedrock" {
+       email                = "spola2@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

Plan: 11 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 11 to add, 0 to change, 0 to destroy.


2. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.36s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_cmu-eats_ch0_anish will be created
+ resource "null_resource" "slack_cmu-eats_ch0_anish" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C043UV30DHA"
          + "user"    = "U07FPJKFB5E"
        }
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

Plan: 1 to add, 0 to change, 0 to destroy.


3. project: openbao dir: tofu/openbao workspace: default

Show Output
Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # vault_identity_group.eats_admins will be created
+ resource "vault_identity_group" "eats_admins" {
+       external_member_entity_ids = false
+       external_member_group_ids  = false
+       external_policies          = false
+       id                         = (known after apply)
+       name                       = "eats-admins"
+       policies                   = [
+           "eats-prod",
        ]
+       type                       = "external"
    }

  # vault_identity_group.eats_members will be created
+ resource "vault_identity_group" "eats_members" {
+       external_member_entity_ids = false
+       external_member_group_ids  = false
+       external_policies          = false
+       id                         = (known after apply)
+       name                       = "eats-members"
+       policies                   = [
+           "eats-dev",
        ]
+       type                       = "external"
    }

  # vault_identity_group_alias.eats_admins will be created
+ resource "vault_identity_group_alias" "eats_admins" {
+       canonical_id   = (known after apply)
+       id             = (known after apply)
+       mount_accessor = "auth_oidc_375ac979"
+       name           = "/projects/eats/admins"
    }

  # vault_identity_group_alias.eats_members will be created
+ resource "vault_identity_group_alias" "eats_members" {
+       canonical_id   = (known after apply)
+       id             = (known after apply)
+       mount_accessor = "auth_oidc_375ac979"
+       name           = "/projects/eats"
    }

  # vault_policy.eats_dev will be created
+ resource "vault_policy" "eats_dev" {
+       allow_overwrite = (known after apply)
+       id              = (known after apply)
+       name            = "eats-dev"
+       policy          = <<-EOT
            path "secret/data/secretspec/eats/dev/*" {
              capabilities = ["create", "read", "update", "delete", "list"]
            }
            
            path "secret/metadata/secretspec/eats/dev/*" {
              capabilities = ["list", "read", "delete"]
            }
            
            path "secret/data/secretspec/eats/preview/*" {
              capabilities = ["create", "read", "update", "delete", "list"]
            }
            
            path "secret/metadata/secretspec/eats/preview/*" {
              capabilities = ["list", "read", "delete"]
            }
        EOT
    }

  # vault_policy.eats_prod will be created
+ resource "vault_policy" "eats_prod" {
+       allow_overwrite = (known after apply)
+       id              = (known after apply)
+       name            = "eats-prod"
+       policy          = <<-EOT
            path "secret/data/secretspec/eats/prod/*" {
              capabilities = ["create", "read", "update", "delete", "list"]
            }
            
            path "secret/metadata/secretspec/eats/prod/*" {
              capabilities = ["list", "read", "delete"]
            }
            
            path "secret/data/secretspec/eats/staging/*" {
              capabilities = ["create", "read", "update", "delete", "list"]
            }
            
            path "secret/metadata/secretspec/eats/staging/*" {
              capabilities = ["list", "read", "delete"]
            }
        EOT
    }

Plan: 6 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p openbao
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p openbao
    

Plan: 6 to add, 0 to change, 0 to destroy.


4. project: posthog dir: tofu/posthog workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.33s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # posthog_project.cmu-eats_eats will be created
+ resource "posthog_project" "cmu-eats_eats" {
+       api_token       = (sensitive value)
+       id              = (known after apply)
+       name            = "eats"
+       organization_id = (known after apply)
+       timezone        = (known after apply)
    }

  # vault_kv_secret_v2.cmu-eats_eats_posthog_host will be created
+ resource "vault_kv_secret_v2" "cmu-eats_eats_posthog_host" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/prod/POSTHOG_HOST"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.cmu-eats_eats_posthog_key will be created
+ resource "vault_kv_secret_v2" "cmu-eats_eats_posthog_key" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/prod/POSTHOG_KEY"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

Plan: 3 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p posthog
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p posthog
    

Plan: 3 to add, 0 to change, 0 to destroy.


5. project: matrix_bridges dir: tofu/matrix_bridges workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on shared package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.36s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # null_resource.slack_bridge_login_cmu_eats will be created
+ resource "null_resource" "slack_bridge_login_cmu_eats" {
+       id       = (known after apply)
+       triggers = {
          + "channel" = "C043UV30DHA"
        }
    }

  # synapse_mautrix_slack_link.cmu_eats will be created
+ resource "synapse_mautrix_slack_link" "cmu_eats" {
+       discord_channel_id       = "1541336561105240134"
+       discord_relay_configured = (known after apply)
+       id                       = (known after apply)
+       room_alias               = (known after apply)
+       room_id                  = (known after apply)
+       slack_bridged            = (known after apply)
+       slack_channel_id         = "C043UV30DHA"
+       team_name                = "CMU Eats"
+       team_slug                = "cmu-eats"
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p matrix_bridges
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p matrix_bridges
    

Plan: 2 to add, 0 to change, 0 to destroy.


6. project: sentry dir: tofu/sentry workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.37s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # sentry_key.cmu-eats_eats will be created
+ resource "sentry_key" "cmu-eats_eats" {
+       dsn                      = (sensitive value)
+       dsn_csp                  = (known after apply)
+       dsn_public               = (known after apply)
+       dsn_secret               = (sensitive value)
+       id                       = (known after apply)
+       javascript_loader_script = (known after apply)
+       name                     = "default"
+       organization             = "scottylabs"
+       project                  = "eats"
+       project_id               = (known after apply)
+       public                   = (known after apply)
+       rate_limit_count         = (known after apply)
+       rate_limit_window        = (known after apply)
+       secret                   = (sensitive value)
    }

  # sentry_project.cmu-eats_eats will be created
+ resource "sentry_project" "cmu-eats_eats" {
+       client_security       = (known after apply)
+       digests_max_delay     = (known after apply)
+       digests_min_delay     = (known after apply)
+       features              = (known after apply)
+       filters               = (known after apply)
+       fingerprinting_rules  = (known after apply)
+       grouping_enhancements = (known after apply)
+       highlight_tags        = (known after apply)
+       id                    = (known after apply)
+       internal_id           = (known after apply)
+       name                  = "eats"
+       organization          = "scottylabs"
+       platform              = "deno"
+       resolve_age           = (known after apply)
+       slug                  = "eats"
+       teams                 = [
+           "cmu-eats",
        ]
    }

  # sentry_team.cmu-eats will be created
+ resource "sentry_team" "cmu-eats" {
+       has_access   = (known after apply)
+       id           = (known after apply)
+       internal_id  = (known after apply)
+       is_member    = (known after apply)
+       is_pending   = (known after apply)
+       name         = "CMU Eats"
+       organization = "scottylabs"
+       slug         = "cmu-eats"
+       team_id      = (known after apply)
    }

  # vault_kv_secret_v2.cmu-eats_eats_sentry_dsn will be created
+ resource "vault_kv_secret_v2" "cmu-eats_eats_sentry_dsn" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/prod/SENTRY_DSN"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

Plan: 4 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p sentry
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p sentry
    

Plan: 4 to add, 0 to change, 0 to destroy.


7. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.28s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # discord_channel_permission.cmu-eats_ch0 will be created
+ resource "discord_channel_permission" "cmu-eats_ch0" {
+       allow        = 1024
+       channel_id   = "1541336561105240134"
+       id           = (known after apply)
+       overwrite_id = (known after apply)
+       type         = "role"
    }

  # discord_member_roles.anish will be updated in-place
! resource "discord_member_roles" "anish" {
        id        = "1353118209024331948:315249559473225729"
        # (2 unchanged attributes hidden)

+       role {
+           has_role = true
+           role_id  = (known after apply)
        }

        # (4 unchanged blocks hidden)
    }

  # discord_role.cmu-eats will be created
+ resource "discord_role" "cmu-eats" {
+       hoist       = false
+       id          = (known after apply)
+       managed     = (known after apply)
+       mentionable = false
+       name        = "CMU Eats"
+       permissions = 0
+       position    = (known after apply)
+       server_id   = "1353118209024331948"
    }

Plan: 2 to add, 1 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

Plan: 2 to add, 1 to change, 0 to destroy.


8. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.32s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # keycloak_group.project_cmu-eats_eats will be created
+ resource "keycloak_group" "project_cmu-eats_eats" {
+       id        = (known after apply)
+       name      = "eats"
+       parent_id = "9e6890c8-50a0-4b82-86b9-91791fffac25"
+       path      = (known after apply)
+       realm_id  = "scottylabs"
    }

  # keycloak_group.project_cmu-eats_eats_admins will be created
+ resource "keycloak_group" "project_cmu-eats_eats_admins" {
+       id        = (known after apply)
+       name      = "admins"
+       parent_id = (known after apply)
+       path      = (known after apply)
+       realm_id  = "scottylabs"
    }

  # keycloak_group_memberships.project_cmu-eats_eats will be created
+ resource "keycloak_group_memberships" "project_cmu-eats_eats" {
+       group_id = (known after apply)
+       id       = (known after apply)
+       members  = [
+           "apallati",
        ]
+       realm_id = "scottylabs"
    }

  # keycloak_group_memberships.project_cmu-eats_eats_admins will be created
+ resource "keycloak_group_memberships" "project_cmu-eats_eats_admins" {
+       group_id = (known after apply)
+       id       = (known after apply)
+       members  = [
+           "apallati",
        ]
+       realm_id = "scottylabs"
    }

  # keycloak_openid_client.eats will be created
+ resource "keycloak_openid_client" "eats" {
+       access_token_lifespan                     = (known after apply)
+       access_type                               = "CONFIDENTIAL"
+       always_display_in_console                 = false
+       backchannel_logout_session_required       = true
+       client_authenticator_type                 = "client-secret"
+       client_id                                 = "eats"
+       client_offline_session_idle_timeout       = (known after apply)
+       client_offline_session_max_lifespan       = (known after apply)
+       client_secret                             = (sensitive value)
+       client_secret_wo                          = (write-only attribute)
+       client_session_idle_timeout               = (known after apply)
+       client_session_max_lifespan               = (known after apply)
+       consent_required                          = (known after apply)
+       direct_access_grants_enabled              = false
+       display_on_consent_screen                 = (known after apply)
+       enabled                                   = true
+       exclude_issuer_from_auth_response         = (known after apply)
+       exclude_session_state_from_auth_response  = (known after apply)
+       frontchannel_logout_enabled               = (known after apply)
+       full_scope_allowed                        = true
+       id                                        = (known after apply)
+       implicit_flow_enabled                     = (known after apply)
+       import                                    = false
+       name                                      = "eats"
+       oauth2_device_authorization_grant_enabled = false
+       oauth2_jwt_authorization_grant_enabled    = false
+       realm_id                                  = "scottylabs"
+       require_dpop_bound_tokens                 = (known after apply)
+       resource_server_id                        = (known after apply)
+       service_account_user_id                   = (known after apply)
+       service_accounts_enabled                  = (known after apply)
+       standard_flow_enabled                     = true
+       standard_token_exchange_enabled           = false
+       use_refresh_tokens                        = true
+       use_refresh_tokens_client_credentials     = false
+       valid_post_logout_redirect_uris           = (known after apply)
+       valid_redirect_uris                       = [
+           "https://oauth.scottylabs.org/oauth2/callback",
        ]
+       web_origins                               = [
+           "+",
        ]
    }

  # keycloak_openid_client.eats_dev will be created
+ resource "keycloak_openid_client" "eats_dev" {
+       access_token_lifespan                     = (known after apply)
+       access_type                               = "CONFIDENTIAL"
+       always_display_in_console                 = false
+       backchannel_logout_session_required       = true
+       client_authenticator_type                 = "client-secret"
+       client_id                                 = "eats-dev"
+       client_offline_session_idle_timeout       = (known after apply)
+       client_offline_session_max_lifespan       = (known after apply)
+       client_secret                             = (sensitive value)
+       client_secret_wo                          = (write-only attribute)
+       client_session_idle_timeout               = (known after apply)
+       client_session_max_lifespan               = (known after apply)
+       consent_required                          = (known after apply)
+       direct_access_grants_enabled              = false
+       display_on_consent_screen                 = (known after apply)
+       enabled                                   = true
+       exclude_issuer_from_auth_response         = (known after apply)
+       exclude_session_state_from_auth_response  = (known after apply)
+       frontchannel_logout_enabled               = (known after apply)
+       full_scope_allowed                        = true
+       id                                        = (known after apply)
+       implicit_flow_enabled                     = (known after apply)
+       import                                    = false
+       name                                      = "eats-dev"
+       oauth2_device_authorization_grant_enabled = false
+       oauth2_jwt_authorization_grant_enabled    = false
+       realm_id                                  = "scottylabs"
+       require_dpop_bound_tokens                 = (known after apply)
+       resource_server_id                        = (known after apply)
+       service_account_user_id                   = (known after apply)
+       service_accounts_enabled                  = (known after apply)
+       standard_flow_enabled                     = true
+       standard_token_exchange_enabled           = false
+       use_refresh_tokens                        = true
+       use_refresh_tokens_client_credentials     = false
+       valid_post_logout_redirect_uris           = (known after apply)
+       valid_redirect_uris                       = [
+           "http://localhost:8090/oauth2/callback",
        ]
+       web_origins                               = [
+           "+",
        ]
    }

  # keycloak_openid_client.eats_staging will be created
+ resource "keycloak_openid_client" "eats_staging" {
+       access_token_lifespan                     = (known after apply)
+       access_type                               = "CONFIDENTIAL"
+       always_display_in_console                 = false
+       backchannel_logout_session_required       = true
+       client_authenticator_type                 = "client-secret"
+       client_id                                 = "eats-staging"
+       client_offline_session_idle_timeout       = (known after apply)
+       client_offline_session_max_lifespan       = (known after apply)
+       client_secret                             = (sensitive value)
+       client_secret_wo                          = (write-only attribute)
+       client_session_idle_timeout               = (known after apply)
+       client_session_max_lifespan               = (known after apply)
+       consent_required                          = (known after apply)
+       direct_access_grants_enabled              = false
+       display_on_consent_screen                 = (known after apply)
+       enabled                                   = true
+       exclude_issuer_from_auth_response         = (known after apply)
+       exclude_session_state_from_auth_response  = (known after apply)
+       frontchannel_logout_enabled               = (known after apply)
+       full_scope_allowed                        = true
+       id                                        = (known after apply)
+       implicit_flow_enabled                     = (known after apply)
+       import                                    = false
+       name                                      = "eats-staging"
+       oauth2_device_authorization_grant_enabled = false
+       oauth2_jwt_authorization_grant_enabled    = false
+       realm_id                                  = "scottylabs"
+       require_dpop_bound_tokens                 = (known after apply)
+       resource_server_id                        = (known after apply)
+       service_account_user_id                   = (known after apply)
+       service_accounts_enabled                  = (known after apply)
+       standard_flow_enabled                     = true
+       standard_token_exchange_enabled           = false
+       use_refresh_tokens                        = true
+       use_refresh_tokens_client_credentials     = false
+       valid_post_logout_redirect_uris           = (known after apply)
+       valid_redirect_uris                       = [
+           "https://oauth.scottylabs.org/oauth2/callback",
        ]
+       web_origins                               = [
+           "+",
        ]
    }

  # keycloak_openid_group_membership_protocol_mapper.eats will be created
+ resource "keycloak_openid_group_membership_protocol_mapper" "eats" {
+       add_to_access_token        = true
+       add_to_id_token            = true
+       add_to_token_introspection = false
+       add_to_userinfo            = true
+       claim_name                 = "groups"
+       client_id                  = (known after apply)
+       full_path                  = true
+       id                         = (known after apply)
+       name                       = "groups"
+       realm_id                   = "scottylabs"
    }

  # keycloak_openid_group_membership_protocol_mapper.eats_dev will be created
+ resource "keycloak_openid_group_membership_protocol_mapper" "eats_dev" {
+       add_to_access_token        = true
+       add_to_id_token            = true
+       add_to_token_introspection = false
+       add_to_userinfo            = true
+       claim_name                 = "groups"
+       client_id                  = (known after apply)
+       full_path                  = true
+       id                         = (known after apply)
+       name                       = "groups"
+       realm_id                   = "scottylabs"
    }

  # keycloak_openid_group_membership_protocol_mapper.eats_staging will be created
+ resource "keycloak_openid_group_membership_protocol_mapper" "eats_staging" {
+       add_to_access_token        = true
+       add_to_id_token            = true
+       add_to_token_introspection = false
+       add_to_userinfo            = true
+       claim_name                 = "groups"
+       client_id                  = (known after apply)
+       full_path                  = true
+       id                         = (known after apply)
+       name                       = "groups"
+       realm_id                   = "scottylabs"
    }

  # vault_kv_secret_v2.eats_dev_keycloak_realm will be created
+ resource "vault_kv_secret_v2" "eats_dev_keycloak_realm" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/dev/KEYCLOAK_REALM"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_dev_keycloak_url will be created
+ resource "vault_kv_secret_v2" "eats_dev_keycloak_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/dev/KEYCLOAK_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_dev_oauth_relay_url will be created
+ resource "vault_kv_secret_v2" "eats_dev_oauth_relay_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/dev/OAUTH_RELAY_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_dev_oidc_client_id will be created
+ resource "vault_kv_secret_v2" "eats_dev_oidc_client_id" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/dev/OIDC_CLIENT_ID"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_dev_oidc_client_secret will be created
+ resource "vault_kv_secret_v2" "eats_dev_oidc_client_secret" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/dev/OIDC_CLIENT_SECRET"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_dev_project_admin_group will be created
+ resource "vault_kv_secret_v2" "eats_dev_project_admin_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/dev/PROJECT_ADMIN_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_dev_project_group will be created
+ resource "vault_kv_secret_v2" "eats_dev_project_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/dev/PROJECT_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_preview_keycloak_realm will be created
+ resource "vault_kv_secret_v2" "eats_preview_keycloak_realm" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/preview/KEYCLOAK_REALM"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_preview_keycloak_url will be created
+ resource "vault_kv_secret_v2" "eats_preview_keycloak_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/preview/KEYCLOAK_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_preview_oauth_relay_url will be created
+ resource "vault_kv_secret_v2" "eats_preview_oauth_relay_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/preview/OAUTH_RELAY_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_preview_oidc_client_id will be created
+ resource "vault_kv_secret_v2" "eats_preview_oidc_client_id" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/preview/OIDC_CLIENT_ID"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_preview_oidc_client_secret will be created
+ resource "vault_kv_secret_v2" "eats_preview_oidc_client_secret" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/preview/OIDC_CLIENT_SECRET"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_preview_project_admin_group will be created
+ resource "vault_kv_secret_v2" "eats_preview_project_admin_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/preview/PROJECT_ADMIN_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_preview_project_group will be created
+ resource "vault_kv_secret_v2" "eats_preview_project_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/preview/PROJECT_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_prod_keycloak_realm will be created
+ resource "vault_kv_secret_v2" "eats_prod_keycloak_realm" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/prod/KEYCLOAK_REALM"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_prod_keycloak_url will be created
+ resource "vault_kv_secret_v2" "eats_prod_keycloak_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/prod/KEYCLOAK_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_prod_oauth_relay_url will be created
+ resource "vault_kv_secret_v2" "eats_prod_oauth_relay_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/prod/OAUTH_RELAY_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_prod_oidc_client_id will be created
+ resource "vault_kv_secret_v2" "eats_prod_oidc_client_id" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/prod/OIDC_CLIENT_ID"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_prod_oidc_client_secret will be created
+ resource "vault_kv_secret_v2" "eats_prod_oidc_client_secret" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/prod/OIDC_CLIENT_SECRET"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_prod_project_admin_group will be created
+ resource "vault_kv_secret_v2" "eats_prod_project_admin_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/prod/PROJECT_ADMIN_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_prod_project_group will be created
+ resource "vault_kv_secret_v2" "eats_prod_project_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/prod/PROJECT_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_staging_keycloak_realm will be created
+ resource "vault_kv_secret_v2" "eats_staging_keycloak_realm" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/staging/KEYCLOAK_REALM"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_staging_keycloak_url will be created
+ resource "vault_kv_secret_v2" "eats_staging_keycloak_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/staging/KEYCLOAK_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_staging_oauth_relay_url will be created
+ resource "vault_kv_secret_v2" "eats_staging_oauth_relay_url" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/staging/OAUTH_RELAY_URL"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_staging_oidc_client_id will be created
+ resource "vault_kv_secret_v2" "eats_staging_oidc_client_id" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/staging/OIDC_CLIENT_ID"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_staging_oidc_client_secret will be created
+ resource "vault_kv_secret_v2" "eats_staging_oidc_client_secret" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/staging/OIDC_CLIENT_SECRET"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_staging_project_admin_group will be created
+ resource "vault_kv_secret_v2" "eats_staging_project_admin_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/staging/PROJECT_ADMIN_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

  # vault_kv_secret_v2.eats_staging_project_group will be created
+ resource "vault_kv_secret_v2" "eats_staging_project_group" {
+       data                = (sensitive value)
+       data_json           = (sensitive value)
+       data_json_wo        = (write-only attribute)
+       delete_all_versions = false
+       disable_read        = false
+       id                  = (known after apply)
+       metadata            = (known after apply)
+       mount               = "secret"
+       name                = "secretspec/eats/staging/PROJECT_GROUP"
+       path                = (known after apply)

+       custom_metadata (known after apply)
    }

Plan: 38 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 38 to add, 0 to change, 0 to destroy.


9. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.41s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create
! update in-place (current -> planned)

OpenTofu will perform the following actions:

  # forgejo_repository.cmu-eats_eats will be created
+ resource "forgejo_repository" "cmu-eats_eats" {
+       allow_manual_merge          = false
+       allow_merge_commits         = true
+       allow_rebase                = true
+       allow_rebase_explicit       = true
+       allow_squash_merge          = true
+       archive_on_destroy          = false
+       archived                    = false
+       auto_init                   = true
+       autodetect_manual_merge     = false
+       avatar_url                  = (known after apply)
+       clone_url                   = (known after apply)
+       created_at                  = (known after apply)
+       default_branch              = "main"
+       default_merge_style         = "merge"
+       empty                       = (known after apply)
+       external_tracker            = (known after apply)
+       external_wiki               = (known after apply)
+       fork                        = (known after apply)
+       forks_count                 = (known after apply)
+       full_name                   = (known after apply)
+       has_actions                 = true
+       has_issues                  = true
+       has_packages                = true
+       has_projects                = true
+       has_pull_requests           = true
+       has_releases                = true
+       has_wiki                    = true
+       html_url                    = (known after apply)
+       id                          = (known after apply)
+       ignore_whitespace_conflicts = false
+       internal                    = (known after apply)
+       internal_tracker            = (known after apply)
+       labels                      = false
+       lfs                         = false
+       milestones                  = false
+       mirror                      = false
+       mirror_interval             = (known after apply)
+       mirror_updated              = (known after apply)
+       name                        = "eats"
+       open_issues_count           = (known after apply)
+       open_pr_counter             = (known after apply)
+       owner                       = "ScottyLabs"
+       parent_id                   = (known after apply)
+       permissions                 = (known after apply)
+       private                     = false
+       release_counter             = (known after apply)
+       size                        = (known after apply)
+       ssh_url                     = (known after apply)
+       stars_count                 = (known after apply)
+       template                    = false
+       trust_model                 = "default"
+       updated_at                  = (known after apply)
+       watchers_count              = (known after apply)
    }

  # forgejo_team.cmu-eats will be created
+ resource "forgejo_team" "cmu-eats" {
+       can_create_org_repo       = (known after apply)
+       description               = "CMU Eats"
+       id                        = (known after apply)
+       includes_all_repositories = (known after apply)
+       name                      = "cmu-eats"
+       organization              = (known after apply)
+       organization_id           = 3
+       permission                = "write"
+       units_map                 = {
          + "repo.code"   = "write"
          + "repo.issues" = "write"
          + "repo.pulls"  = "write"
        }
    }

  # forgejo_team.eats_leads will be created
+ resource "forgejo_team" "eats_leads" {
+       can_create_org_repo       = false
+       description               = "CMU Eats leads"
+       id                        = (known after apply)
+       includes_all_repositories = false
+       name                      = "eats-leads"
+       organization              = (known after apply)
+       organization_id           = 3
+       permission                = "admin"
+       units_map                 = {
          + "repo.actions"    = "admin"
          + "repo.code"       = "admin"
          + "repo.ext_issues" = "admin"
          + "repo.ext_wiki"   = "admin"
          + "repo.issues"     = "admin"
          + "repo.packages"   = "admin"
          + "repo.projects"   = "admin"
          + "repo.pulls"      = "admin"
          + "repo.releases"   = "admin"
          + "repo.wiki"       = "admin"
        }
    }

  # forgejo_team_member.cmu-eats_anish will be created
+ resource "forgejo_team_member" "cmu-eats_anish" {
+       team_id = (known after apply)
+       user    = "anish"
    }

  # forgejo_team_member.eats_leads_anish will be created
+ resource "forgejo_team_member" "eats_leads_anish" {
+       team_id = (known after apply)
+       user    = "anish"
    }

  # forgejo_team_repository.cmu-eats_eats will be created
+ resource "forgejo_team_repository" "cmu-eats_eats" {
+       owner      = "ScottyLabs"
+       repository = "eats"
+       team_id    = (known after apply)
    }

  # forgejo_team_repository.eats_leads_eats will be created
+ resource "forgejo_team_repository" "eats_leads_eats" {
+       owner      = "ScottyLabs"
+       repository = "eats"
+       team_id    = (known after apply)
    }

  # github_actions_repository_permissions.cmu-eats_eats_actions will be created
+ resource "github_actions_repository_permissions" "cmu-eats_eats_actions" {
+       enabled              = false
+       id                   = (known after apply)
+       repository           = "eats"
+       sha_pinning_required = (known after apply)
    }

  # github_repository.cmu-eats_eats will be created
+ resource "github_repository" "cmu-eats_eats" {
+       allow_auto_merge                        = false
+       allow_forking                           = (known after apply)
+       allow_merge_commit                      = true
+       allow_rebase_merge                      = true
+       allow_squash_merge                      = true
+       archived                                = false
+       default_branch                          = (known after apply)
+       delete_branch_on_merge                  = false
+       description                             = "READ-ONLY MIRROR"
+       etag                                    = (known after apply)
+       fork                                    = (known after apply)
+       full_name                               = (known after apply)
+       git_clone_url                           = (known after apply)
+       has_discussions                         = false
+       has_issues                              = false
+       has_projects                            = false
+       has_wiki                                = false
+       homepage_url                            = "https://git.cmu.dev/ScottyLabs/eats"
+       html_url                                = (known after apply)
+       http_clone_url                          = (known after apply)
+       id                                      = (known after apply)
+       ignore_vulnerability_alerts_during_read = false
+       merge_commit_message                    = "PR_TITLE"
+       merge_commit_title                      = "MERGE_MESSAGE"
+       name                                    = "eats"
+       node_id                                 = (known after apply)
+       primary_language                        = (known after apply)
+       private                                 = (known after apply)
+       repo_id                                 = (known after apply)
+       source_owner                            = (known after apply)
+       source_repo                             = (known after apply)
+       squash_merge_commit_message             = "COMMIT_MESSAGES"
+       squash_merge_commit_title               = "COMMIT_OR_PR_TITLE"
+       ssh_clone_url                           = (known after apply)
+       svn_url                                 = (known after apply)
+       topics                                  = (known after apply)
+       visibility                              = "public"
+       vulnerability_alerts                    = (known after apply)
+       web_commit_signoff_required             = (known after apply)

+       security_and_analysis (known after apply)
    }

  # github_repository.slai_cmugpt_agent will be updated in-place
! resource "github_repository" "slai_cmugpt_agent" {
        id                                      = "cmugpt-agent"
        name                                    = "cmugpt-agent"
        # (39 unchanged attributes hidden)

-       template {
-           include_all_branches = false -> null
-           owner                = "ScottyLabs" -> null
-           repository           = "python-template" -> null
        }

        # (1 unchanged block hidden)
    }

  # github_repository.slai_cmugpt_surface will be updated in-place
! resource "github_repository" "slai_cmugpt_surface" {
        id                                      = "cmugpt-surface"
        name                                    = "cmugpt-surface"
        # (39 unchanged attributes hidden)

-       template {
-           include_all_branches = false -> null
-           owner                = "scottylabs-labrador" -> null
-           repository           = "ScottyStack" -> null
        }

        # (1 unchanged block hidden)
    }

  # github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key will be created
+ resource "github_repository_deploy_key" "cmu-eats_eats_mirror_deploy_key" {
+       etag       = (known after apply)
+       id         = (known after apply)
+       key        = (known after apply)
+       read_only  = false
+       repository = "eats"
+       title      = "Forgejo mirror"
    }

  # github_repository_ruleset.cbp_bus_sign_ruleset will be updated in-place
! resource "github_repository_ruleset" "cbp_bus_sign_ruleset" {
        id          = "15236550"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665342
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.cbp_dalmatian_ruleset will be updated in-place
! resource "github_repository_ruleset" "cbp_dalmatian_ruleset" {
        id          = "17238915"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665367
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.cbp_discord_verify_ruleset will be updated in-place
! resource "github_repository_ruleset" "cbp_discord_verify_ruleset" {
        id          = "11449154"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665340
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.cbp_groupme_mirror_ruleset will be updated in-place
! resource "github_repository_ruleset" "cbp_groupme_mirror_ruleset" {
        id          = "12503296"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665366
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.cmu-cal_cal_ruleset will be updated in-place
! resource "github_repository_ruleset" "cmu-cal_cal_ruleset" {
        id          = "18919281"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665344
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.cmu-courses_courses_ruleset will be updated in-place
! resource "github_repository_ruleset" "cmu-courses_courses_ruleset" {
        id          = "15855161"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665327
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.cmu-courses_internet_archive_ruleset will be updated in-place
! resource "github_repository_ruleset" "cmu-courses_internet_archive_ruleset" {
        id          = "11988074"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665356
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.cmu-eats_eats_ruleset will be created
+ resource "github_repository_ruleset" "cmu-eats_eats_ruleset" {
+       enforcement = "active"
+       etag        = (known after apply)
+       id          = (known after apply)
+       name        = "Default"
+       node_id     = (known after apply)
+       repository  = "eats"
+       ruleset_id  = (known after apply)
+       target      = "branch"

+       bypass_actors {
+           actor_id    = (known after apply)
+           actor_type  = "DeployKey"
+           bypass_mode = "always"
        }

+       conditions {
+           ref_name {
+               exclude = []
+               include = [
+                   "~ALL",
                ]
            }
        }

+       rules {
+           creation                      = true
+           deletion                      = true
+           non_fast_forward              = true
+           update                        = true
+           update_allows_fetch_and_merge = false

+           pull_request {
+               allowed_merge_methods             = (known after apply)
+               dismiss_stale_reviews_on_push     = false
+               require_code_owner_review         = false
+               require_last_push_approval        = false
+               required_approving_review_count   = 1
+               required_review_thread_resolution = false
            }
        }
    }

  # github_repository_ruleset.cmu-housing_housing_ruleset will be updated in-place
! resource "github_repository_ruleset" "cmu-housing_housing_ruleset" {
        id          = "17421604"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665369
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.cmu-maps_maps_ruleset will be updated in-place
! resource "github_repository_ruleset" "cmu-maps_maps_ruleset" {
        id          = "18864961"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665325
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.devops_collie_ruleset will be updated in-place
! resource "github_repository_ruleset" "devops_collie_ruleset" {
        id          = "18865204"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665332
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.devops_documentation_ruleset will be updated in-place
! resource "github_repository_ruleset" "devops_documentation_ruleset" {
        id          = "17079368"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665362
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.devops_governance_ruleset will be updated in-place
! resource "github_repository_ruleset" "devops_governance_ruleset" {
        id          = "14488792"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665346
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.devops_infrastructure_ruleset will be updated in-place
! resource "github_repository_ruleset" "devops_infrastructure_ruleset" {
        id          = "11275118"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665343
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.devops_kennel_ruleset will be updated in-place
! resource "github_repository_ruleset" "devops_kennel_ruleset" {
        id          = "13193453"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665328
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.devops_link_shortener_ruleset will be updated in-place
! resource "github_repository_ruleset" "devops_link_shortener_ruleset" {
        id          = "17911647"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665345
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset will be updated in-place
! resource "github_repository_ruleset" "lost-and-found_lost_and_found_v2_ruleset" {
        id          = "18920618"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665337
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.quest_quest_ruleset will be updated in-place
! resource "github_repository_ruleset" "quest_quest_ruleset" {
        id          = "15750052"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665323
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.slai_cmugpt_agent_ruleset will be updated in-place
! resource "github_repository_ruleset" "slai_cmugpt_agent_ruleset" {
        id          = "16476661"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665360
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.slai_cmugpt_sms_surface_ruleset will be updated in-place
! resource "github_repository_ruleset" "slai_cmugpt_sms_surface_ruleset" {
        id          = "17653622"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665365
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.slai_cmugpt_surface_ruleset will be updated in-place
! resource "github_repository_ruleset" "slai_cmugpt_surface_ruleset" {
        id          = "16476662"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665348
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.slai_mcp_server_ruleset will be updated in-place
! resource "github_repository_ruleset" "slai_mcp_server_ruleset" {
        id          = "17082154"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665358
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.tartan-vote_tartan_vote_ruleset will be updated in-place
! resource "github_repository_ruleset" "tartan-vote_tartan_vote_ruleset" {
        id          = "16848966"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665350
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.terrier_terrier_ruleset will be updated in-place
! resource "github_repository_ruleset" "terrier_terrier_ruleset" {
        id          = "12671003"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665339
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.uia_components_ruleset will be updated in-place
! resource "github_repository_ruleset" "uia_components_ruleset" {
        id          = "16366085"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665353
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_repository_ruleset.uia_scottylabs_org_ruleset will be updated in-place
! resource "github_repository_ruleset" "uia_scottylabs_org_ruleset" {
        id          = "18112214"
        name        = "Default"
        # (6 unchanged attributes hidden)

!       bypass_actors {
!           actor_id    = 0 -> 159665361
            # (2 unchanged attributes hidden)
        }

        # (2 unchanged blocks hidden)
    }

  # github_team.cmu-eats will be created
+ resource "github_team" "cmu-eats" {
+       create_default_maintainer = false
+       etag                      = (known after apply)
+       id                        = (known after apply)
+       members_count             = (known after apply)
+       name                      = "CMU Eats"
+       node_id                   = (known after apply)
+       notification_setting      = "notifications_enabled"
+       parent_team_read_id       = (known after apply)
+       parent_team_read_slug     = (known after apply)
+       privacy                   = "closed"
+       slug                      = (known after apply)
    }

  # github_team_membership.cmu-eats_anish will be created
+ resource "github_team_membership" "cmu-eats_anish" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "maintainer"
+       team_id  = (known after apply)
+       username = "ap-1"
    }

  # github_team_membership.quest_brightz will be created
+ resource "github_team_membership" "quest_brightz" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "member"
+       team_id  = "17330197"
+       username = "bright-zheng"
    }

  # restapi_object.cmu-eats_eats_kennel_webhook will be created
+ resource "restapi_object" "cmu-eats_eats_kennel_webhook" {
+       api_data        = (known after apply)
+       api_response    = (known after apply)
+       create_response = (known after apply)
+       data            = (sensitive value)
+       id              = (known after apply)
+       id_attribute    = "id"
+       path            = "/api/v1/repos/ScottyLabs/eats/hooks"
+       update_method   = "PATCH"
    }

  # restapi_object.cmu-eats_eats_push_mirror will be created
+ resource "restapi_object" "cmu-eats_eats_push_mirror" {
+       api_data        = (known after apply)
+       api_response    = (known after apply)
+       create_response = (known after apply)
+       data            = jsonencode(
            {
+               interval       = "8h0m0s"
+               remote_address = "git@github.com:ScottyLabs/eats.git"
+               sync_on_commit = true
+               use_ssh        = true
            }
        )
+       id              = (known after apply)
+       id_attribute    = "remote_name"
+       path            = "/api/v1/repos/ScottyLabs/eats/push_mirrors"
    }

Plan: 16 to add, 27 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 16 to add, 27 to change, 0 to destroy.


Plan Summary

9 projects, 9 with changes, 0 with no changes, 0 failed

  • To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 9 projects: 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `openbao` dir: `tofu/openbao` workspace: `default` 1. project: `posthog` dir: `tofu/posthog` workspace: `default` 1. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` 1. project: `sentry` dir: `tofu/sentry` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.35s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ data.external.identity_anish: Reading... data.external.identity_thesuperrl: Reading... data.external.identity_maybe_yiyi: Reading... data.external.identity_liamwalk: Reading... data.external.identity_tentype: Reading... data.external.identity_brightz: Reading... data.external.identity_jc_965: Reading... data.external.identity_riztofu: Reading... data.external.identity_kechezon: Reading... data.external.identity_xboxbedrock: Reading... data.external.identity_brightz: Read complete after 0s [id=-] data.external.identity_kothurvidur: Reading... data.external.identity_liamwalk: Read complete after 0s [id=-] data.external.identity_jiyas: Reading... data.external.identity_tentype: Read complete after 0s [id=-] data.external.identity_dkou: Reading... data.external.identity_jc_965: Read complete after 0s [id=-] data.external.identity_rzhu3: Reading... data.external.identity_riztofu: Read complete after 0s [id=-] data.external.identity_maybe_yiyi: Read complete after 0s [id=-] data.external.identity_krishsax: Reading... data.external.identity_anish: Read complete after 0s [id=-] data.external.identity_xboxbedrock: Read complete after 0s [id=-] data.external.identity_thesuperrl: Read complete after 0s [id=-] data.external.identity_kechezon: Read complete after 0s [id=-] data.external.identity_jiyas: Read complete after 0s [id=-] data.external.identity_kothurvidur: Read complete after 0s [id=-] data.external.identity_rzhu3: Read complete after 0s [id=-] data.external.identity_krishsax: Read complete after 0s [id=-] data.external.identity_dkou: Read complete after 0s [id=-] OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # googleplay_app_iam.quest_brightz will be created + resource "googleplay_app_iam" "quest_brightz" { + app_id = "4972435805279727996" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "brightz@andrew.cmu.edu" } # googleplay_app_iam.quest_kechezon will be created + resource "googleplay_app_iam" "quest_kechezon" { + app_id = "4972435805279727996" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "kechezon@andrew.cmu.edu" } # googleplay_app_iam.quest_xboxbedrock will be created + resource "googleplay_app_iam" "quest_xboxbedrock" { + app_id = "4972435805279727996" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_APP_CONTENT", + "CAN_MANAGE_PUBLIC_APKS", + "CAN_MANAGE_PUBLIC_LISTING", + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_REPLY_TO_REVIEWS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "spola2@andrew.cmu.edu" } # googleplay_app_iam.terrier_dkou will be created + resource "googleplay_app_iam" "terrier_dkou" { + app_id = "4975152612825535266" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "dkou@andrew.cmu.edu" } # googleplay_app_iam.terrier_kothurvidur will be created + resource "googleplay_app_iam" "terrier_kothurvidur" { + app_id = "4975152612825535266" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "vkothur@andrew.cmu.edu" } # googleplay_user.anish will be created + resource "googleplay_user" "anish" { + email = "apallati@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_MANAGE_PERMISSIONS_GLOBAL", ] + id = (known after apply) + name = (known after apply) } # googleplay_user.brightz will be created + resource "googleplay_user" "brightz" { + email = "brightz@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } # googleplay_user.dkou will be created + resource "googleplay_user" "dkou" { + email = "dkou@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } # googleplay_user.kechezon will be created + resource "googleplay_user" "kechezon" { + email = "kechezon@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } # googleplay_user.kothurvidur will be created + resource "googleplay_user" "kothurvidur" { + email = "vkothur@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } # googleplay_user.xboxbedrock will be created + resource "googleplay_user" "xboxbedrock" { + email = "spola2@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } Plan: 11 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 11 to add, 0 to change, 0 to destroy. --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.36s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_cmu-eats_ch0_anish will be created + resource "null_resource" "slack_cmu-eats_ch0_anish" { + id = (known after apply) + triggers = { + "channel" = "C043UV30DHA" + "user" = "U07FPJKFB5E" } } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### 3. project: `openbao` dir: `tofu/openbao` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # vault_identity_group.eats_admins will be created + resource "vault_identity_group" "eats_admins" { + external_member_entity_ids = false + external_member_group_ids = false + external_policies = false + id = (known after apply) + name = "eats-admins" + policies = [ + "eats-prod", ] + type = "external" } # vault_identity_group.eats_members will be created + resource "vault_identity_group" "eats_members" { + external_member_entity_ids = false + external_member_group_ids = false + external_policies = false + id = (known after apply) + name = "eats-members" + policies = [ + "eats-dev", ] + type = "external" } # vault_identity_group_alias.eats_admins will be created + resource "vault_identity_group_alias" "eats_admins" { + canonical_id = (known after apply) + id = (known after apply) + mount_accessor = "auth_oidc_375ac979" + name = "/projects/eats/admins" } # vault_identity_group_alias.eats_members will be created + resource "vault_identity_group_alias" "eats_members" { + canonical_id = (known after apply) + id = (known after apply) + mount_accessor = "auth_oidc_375ac979" + name = "/projects/eats" } # vault_policy.eats_dev will be created + resource "vault_policy" "eats_dev" { + allow_overwrite = (known after apply) + id = (known after apply) + name = "eats-dev" + policy = <<-EOT path "secret/data/secretspec/eats/dev/*" { capabilities = ["create", "read", "update", "delete", "list"] } path "secret/metadata/secretspec/eats/dev/*" { capabilities = ["list", "read", "delete"] } path "secret/data/secretspec/eats/preview/*" { capabilities = ["create", "read", "update", "delete", "list"] } path "secret/metadata/secretspec/eats/preview/*" { capabilities = ["list", "read", "delete"] } EOT } # vault_policy.eats_prod will be created + resource "vault_policy" "eats_prod" { + allow_overwrite = (known after apply) + id = (known after apply) + name = "eats-prod" + policy = <<-EOT path "secret/data/secretspec/eats/prod/*" { capabilities = ["create", "read", "update", "delete", "list"] } path "secret/metadata/secretspec/eats/prod/*" { capabilities = ["list", "read", "delete"] } path "secret/data/secretspec/eats/staging/*" { capabilities = ["create", "read", "update", "delete", "list"] } path "secret/metadata/secretspec/eats/staging/*" { capabilities = ["list", "read", "delete"] } EOT } Plan: 6 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p openbao ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fopenbao%252Fdefault%252Fopenbao) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p openbao ``` Plan: 6 to add, 0 to change, 0 to destroy. --- ### 4. project: `posthog` dir: `tofu/posthog` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.33s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # posthog_project.cmu-eats_eats will be created + resource "posthog_project" "cmu-eats_eats" { + api_token = (sensitive value) + id = (known after apply) + name = "eats" + organization_id = (known after apply) + timezone = (known after apply) } # vault_kv_secret_v2.cmu-eats_eats_posthog_host will be created + resource "vault_kv_secret_v2" "cmu-eats_eats_posthog_host" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/prod/POSTHOG_HOST" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.cmu-eats_eats_posthog_key will be created + resource "vault_kv_secret_v2" "cmu-eats_eats_posthog_key" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/prod/POSTHOG_KEY" + path = (known after apply) + custom_metadata (known after apply) } Plan: 3 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p posthog ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fposthog%252Fdefault%252Fposthog) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p posthog ``` Plan: 3 to add, 0 to change, 0 to destroy. --- ### 5. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on shared package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.36s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # null_resource.slack_bridge_login_cmu_eats will be created + resource "null_resource" "slack_bridge_login_cmu_eats" { + id = (known after apply) + triggers = { + "channel" = "C043UV30DHA" } } # synapse_mautrix_slack_link.cmu_eats will be created + resource "synapse_mautrix_slack_link" "cmu_eats" { + discord_channel_id = "1541336561105240134" + discord_relay_configured = (known after apply) + id = (known after apply) + room_alias = (known after apply) + room_id = (known after apply) + slack_bridged = (known after apply) + slack_channel_id = "C043UV30DHA" + team_name = "CMU Eats" + team_slug = "cmu-eats" } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p matrix_bridges ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fmatrix_bridges%252Fdefault%252Fmatrix_bridges) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p matrix_bridges ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### 6. project: `sentry` dir: `tofu/sentry` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.37s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # sentry_key.cmu-eats_eats will be created + resource "sentry_key" "cmu-eats_eats" { + dsn = (sensitive value) + dsn_csp = (known after apply) + dsn_public = (known after apply) + dsn_secret = (sensitive value) + id = (known after apply) + javascript_loader_script = (known after apply) + name = "default" + organization = "scottylabs" + project = "eats" + project_id = (known after apply) + public = (known after apply) + rate_limit_count = (known after apply) + rate_limit_window = (known after apply) + secret = (sensitive value) } # sentry_project.cmu-eats_eats will be created + resource "sentry_project" "cmu-eats_eats" { + client_security = (known after apply) + digests_max_delay = (known after apply) + digests_min_delay = (known after apply) + features = (known after apply) + filters = (known after apply) + fingerprinting_rules = (known after apply) + grouping_enhancements = (known after apply) + highlight_tags = (known after apply) + id = (known after apply) + internal_id = (known after apply) + name = "eats" + organization = "scottylabs" + platform = "deno" + resolve_age = (known after apply) + slug = "eats" + teams = [ + "cmu-eats", ] } # sentry_team.cmu-eats will be created + resource "sentry_team" "cmu-eats" { + has_access = (known after apply) + id = (known after apply) + internal_id = (known after apply) + is_member = (known after apply) + is_pending = (known after apply) + name = "CMU Eats" + organization = "scottylabs" + slug = "cmu-eats" + team_id = (known after apply) } # vault_kv_secret_v2.cmu-eats_eats_sentry_dsn will be created + resource "vault_kv_secret_v2" "cmu-eats_eats_sentry_dsn" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/prod/SENTRY_DSN" + path = (known after apply) + custom_metadata (known after apply) } Plan: 4 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p sentry ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fsentry%252Fdefault%252Fsentry) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p sentry ``` Plan: 4 to add, 0 to change, 0 to destroy. --- ### 7. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.28s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # discord_channel_permission.cmu-eats_ch0 will be created + resource "discord_channel_permission" "cmu-eats_ch0" { + allow = 1024 + channel_id = "1541336561105240134" + id = (known after apply) + overwrite_id = (known after apply) + type = "role" } # discord_member_roles.anish will be updated in-place ! resource "discord_member_roles" "anish" { id = "1353118209024331948:315249559473225729" # (2 unchanged attributes hidden) + role { + has_role = true + role_id = (known after apply) } # (4 unchanged blocks hidden) } # discord_role.cmu-eats will be created + resource "discord_role" "cmu-eats" { + hoist = false + id = (known after apply) + managed = (known after apply) + mentionable = false + name = "CMU Eats" + permissions = 0 + position = (known after apply) + server_id = "1353118209024331948" } Plan: 2 to add, 1 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` Plan: 2 to add, 1 to change, 0 to destroy. --- ### 8. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.32s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # keycloak_group.project_cmu-eats_eats will be created + resource "keycloak_group" "project_cmu-eats_eats" { + id = (known after apply) + name = "eats" + parent_id = "9e6890c8-50a0-4b82-86b9-91791fffac25" + path = (known after apply) + realm_id = "scottylabs" } # keycloak_group.project_cmu-eats_eats_admins will be created + resource "keycloak_group" "project_cmu-eats_eats_admins" { + id = (known after apply) + name = "admins" + parent_id = (known after apply) + path = (known after apply) + realm_id = "scottylabs" } # keycloak_group_memberships.project_cmu-eats_eats will be created + resource "keycloak_group_memberships" "project_cmu-eats_eats" { + group_id = (known after apply) + id = (known after apply) + members = [ + "apallati", ] + realm_id = "scottylabs" } # keycloak_group_memberships.project_cmu-eats_eats_admins will be created + resource "keycloak_group_memberships" "project_cmu-eats_eats_admins" { + group_id = (known after apply) + id = (known after apply) + members = [ + "apallati", ] + realm_id = "scottylabs" } # keycloak_openid_client.eats will be created + resource "keycloak_openid_client" "eats" { + access_token_lifespan = (known after apply) + access_type = "CONFIDENTIAL" + always_display_in_console = false + backchannel_logout_session_required = true + client_authenticator_type = "client-secret" + client_id = "eats" + client_offline_session_idle_timeout = (known after apply) + client_offline_session_max_lifespan = (known after apply) + client_secret = (sensitive value) + client_secret_wo = (write-only attribute) + client_session_idle_timeout = (known after apply) + client_session_max_lifespan = (known after apply) + consent_required = (known after apply) + direct_access_grants_enabled = false + display_on_consent_screen = (known after apply) + enabled = true + exclude_issuer_from_auth_response = (known after apply) + exclude_session_state_from_auth_response = (known after apply) + frontchannel_logout_enabled = (known after apply) + full_scope_allowed = true + id = (known after apply) + implicit_flow_enabled = (known after apply) + import = false + name = "eats" + oauth2_device_authorization_grant_enabled = false + oauth2_jwt_authorization_grant_enabled = false + realm_id = "scottylabs" + require_dpop_bound_tokens = (known after apply) + resource_server_id = (known after apply) + service_account_user_id = (known after apply) + service_accounts_enabled = (known after apply) + standard_flow_enabled = true + standard_token_exchange_enabled = false + use_refresh_tokens = true + use_refresh_tokens_client_credentials = false + valid_post_logout_redirect_uris = (known after apply) + valid_redirect_uris = [ + "https://oauth.scottylabs.org/oauth2/callback", ] + web_origins = [ + "+", ] } # keycloak_openid_client.eats_dev will be created + resource "keycloak_openid_client" "eats_dev" { + access_token_lifespan = (known after apply) + access_type = "CONFIDENTIAL" + always_display_in_console = false + backchannel_logout_session_required = true + client_authenticator_type = "client-secret" + client_id = "eats-dev" + client_offline_session_idle_timeout = (known after apply) + client_offline_session_max_lifespan = (known after apply) + client_secret = (sensitive value) + client_secret_wo = (write-only attribute) + client_session_idle_timeout = (known after apply) + client_session_max_lifespan = (known after apply) + consent_required = (known after apply) + direct_access_grants_enabled = false + display_on_consent_screen = (known after apply) + enabled = true + exclude_issuer_from_auth_response = (known after apply) + exclude_session_state_from_auth_response = (known after apply) + frontchannel_logout_enabled = (known after apply) + full_scope_allowed = true + id = (known after apply) + implicit_flow_enabled = (known after apply) + import = false + name = "eats-dev" + oauth2_device_authorization_grant_enabled = false + oauth2_jwt_authorization_grant_enabled = false + realm_id = "scottylabs" + require_dpop_bound_tokens = (known after apply) + resource_server_id = (known after apply) + service_account_user_id = (known after apply) + service_accounts_enabled = (known after apply) + standard_flow_enabled = true + standard_token_exchange_enabled = false + use_refresh_tokens = true + use_refresh_tokens_client_credentials = false + valid_post_logout_redirect_uris = (known after apply) + valid_redirect_uris = [ + "http://localhost:8090/oauth2/callback", ] + web_origins = [ + "+", ] } # keycloak_openid_client.eats_staging will be created + resource "keycloak_openid_client" "eats_staging" { + access_token_lifespan = (known after apply) + access_type = "CONFIDENTIAL" + always_display_in_console = false + backchannel_logout_session_required = true + client_authenticator_type = "client-secret" + client_id = "eats-staging" + client_offline_session_idle_timeout = (known after apply) + client_offline_session_max_lifespan = (known after apply) + client_secret = (sensitive value) + client_secret_wo = (write-only attribute) + client_session_idle_timeout = (known after apply) + client_session_max_lifespan = (known after apply) + consent_required = (known after apply) + direct_access_grants_enabled = false + display_on_consent_screen = (known after apply) + enabled = true + exclude_issuer_from_auth_response = (known after apply) + exclude_session_state_from_auth_response = (known after apply) + frontchannel_logout_enabled = (known after apply) + full_scope_allowed = true + id = (known after apply) + implicit_flow_enabled = (known after apply) + import = false + name = "eats-staging" + oauth2_device_authorization_grant_enabled = false + oauth2_jwt_authorization_grant_enabled = false + realm_id = "scottylabs" + require_dpop_bound_tokens = (known after apply) + resource_server_id = (known after apply) + service_account_user_id = (known after apply) + service_accounts_enabled = (known after apply) + standard_flow_enabled = true + standard_token_exchange_enabled = false + use_refresh_tokens = true + use_refresh_tokens_client_credentials = false + valid_post_logout_redirect_uris = (known after apply) + valid_redirect_uris = [ + "https://oauth.scottylabs.org/oauth2/callback", ] + web_origins = [ + "+", ] } # keycloak_openid_group_membership_protocol_mapper.eats will be created + resource "keycloak_openid_group_membership_protocol_mapper" "eats" { + add_to_access_token = true + add_to_id_token = true + add_to_token_introspection = false + add_to_userinfo = true + claim_name = "groups" + client_id = (known after apply) + full_path = true + id = (known after apply) + name = "groups" + realm_id = "scottylabs" } # keycloak_openid_group_membership_protocol_mapper.eats_dev will be created + resource "keycloak_openid_group_membership_protocol_mapper" "eats_dev" { + add_to_access_token = true + add_to_id_token = true + add_to_token_introspection = false + add_to_userinfo = true + claim_name = "groups" + client_id = (known after apply) + full_path = true + id = (known after apply) + name = "groups" + realm_id = "scottylabs" } # keycloak_openid_group_membership_protocol_mapper.eats_staging will be created + resource "keycloak_openid_group_membership_protocol_mapper" "eats_staging" { + add_to_access_token = true + add_to_id_token = true + add_to_token_introspection = false + add_to_userinfo = true + claim_name = "groups" + client_id = (known after apply) + full_path = true + id = (known after apply) + name = "groups" + realm_id = "scottylabs" } # vault_kv_secret_v2.eats_dev_keycloak_realm will be created + resource "vault_kv_secret_v2" "eats_dev_keycloak_realm" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/dev/KEYCLOAK_REALM" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_dev_keycloak_url will be created + resource "vault_kv_secret_v2" "eats_dev_keycloak_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/dev/KEYCLOAK_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_dev_oauth_relay_url will be created + resource "vault_kv_secret_v2" "eats_dev_oauth_relay_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/dev/OAUTH_RELAY_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_dev_oidc_client_id will be created + resource "vault_kv_secret_v2" "eats_dev_oidc_client_id" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/dev/OIDC_CLIENT_ID" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_dev_oidc_client_secret will be created + resource "vault_kv_secret_v2" "eats_dev_oidc_client_secret" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/dev/OIDC_CLIENT_SECRET" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_dev_project_admin_group will be created + resource "vault_kv_secret_v2" "eats_dev_project_admin_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/dev/PROJECT_ADMIN_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_dev_project_group will be created + resource "vault_kv_secret_v2" "eats_dev_project_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/dev/PROJECT_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_preview_keycloak_realm will be created + resource "vault_kv_secret_v2" "eats_preview_keycloak_realm" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/preview/KEYCLOAK_REALM" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_preview_keycloak_url will be created + resource "vault_kv_secret_v2" "eats_preview_keycloak_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/preview/KEYCLOAK_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_preview_oauth_relay_url will be created + resource "vault_kv_secret_v2" "eats_preview_oauth_relay_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/preview/OAUTH_RELAY_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_preview_oidc_client_id will be created + resource "vault_kv_secret_v2" "eats_preview_oidc_client_id" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/preview/OIDC_CLIENT_ID" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_preview_oidc_client_secret will be created + resource "vault_kv_secret_v2" "eats_preview_oidc_client_secret" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/preview/OIDC_CLIENT_SECRET" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_preview_project_admin_group will be created + resource "vault_kv_secret_v2" "eats_preview_project_admin_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/preview/PROJECT_ADMIN_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_preview_project_group will be created + resource "vault_kv_secret_v2" "eats_preview_project_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/preview/PROJECT_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_prod_keycloak_realm will be created + resource "vault_kv_secret_v2" "eats_prod_keycloak_realm" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/prod/KEYCLOAK_REALM" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_prod_keycloak_url will be created + resource "vault_kv_secret_v2" "eats_prod_keycloak_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/prod/KEYCLOAK_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_prod_oauth_relay_url will be created + resource "vault_kv_secret_v2" "eats_prod_oauth_relay_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/prod/OAUTH_RELAY_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_prod_oidc_client_id will be created + resource "vault_kv_secret_v2" "eats_prod_oidc_client_id" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/prod/OIDC_CLIENT_ID" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_prod_oidc_client_secret will be created + resource "vault_kv_secret_v2" "eats_prod_oidc_client_secret" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/prod/OIDC_CLIENT_SECRET" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_prod_project_admin_group will be created + resource "vault_kv_secret_v2" "eats_prod_project_admin_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/prod/PROJECT_ADMIN_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_prod_project_group will be created + resource "vault_kv_secret_v2" "eats_prod_project_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/prod/PROJECT_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_staging_keycloak_realm will be created + resource "vault_kv_secret_v2" "eats_staging_keycloak_realm" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/staging/KEYCLOAK_REALM" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_staging_keycloak_url will be created + resource "vault_kv_secret_v2" "eats_staging_keycloak_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/staging/KEYCLOAK_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_staging_oauth_relay_url will be created + resource "vault_kv_secret_v2" "eats_staging_oauth_relay_url" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/staging/OAUTH_RELAY_URL" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_staging_oidc_client_id will be created + resource "vault_kv_secret_v2" "eats_staging_oidc_client_id" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/staging/OIDC_CLIENT_ID" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_staging_oidc_client_secret will be created + resource "vault_kv_secret_v2" "eats_staging_oidc_client_secret" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/staging/OIDC_CLIENT_SECRET" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_staging_project_admin_group will be created + resource "vault_kv_secret_v2" "eats_staging_project_admin_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/staging/PROJECT_ADMIN_GROUP" + path = (known after apply) + custom_metadata (known after apply) } # vault_kv_secret_v2.eats_staging_project_group will be created + resource "vault_kv_secret_v2" "eats_staging_project_group" { + data = (sensitive value) + data_json = (sensitive value) + data_json_wo = (write-only attribute) + delete_all_versions = false + disable_read = false + id = (known after apply) + metadata = (known after apply) + mount = "secret" + name = "secretspec/eats/staging/PROJECT_GROUP" + path = (known after apply) + custom_metadata (known after apply) } Plan: 38 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 38 to add, 0 to change, 0 to destroy. --- ### 9. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.41s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create ! update in-place (current -> planned) OpenTofu will perform the following actions: # forgejo_repository.cmu-eats_eats will be created + resource "forgejo_repository" "cmu-eats_eats" { + allow_manual_merge = false + allow_merge_commits = true + allow_rebase = true + allow_rebase_explicit = true + allow_squash_merge = true + archive_on_destroy = false + archived = false + auto_init = true + autodetect_manual_merge = false + avatar_url = (known after apply) + clone_url = (known after apply) + created_at = (known after apply) + default_branch = "main" + default_merge_style = "merge" + empty = (known after apply) + external_tracker = (known after apply) + external_wiki = (known after apply) + fork = (known after apply) + forks_count = (known after apply) + full_name = (known after apply) + has_actions = true + has_issues = true + has_packages = true + has_projects = true + has_pull_requests = true + has_releases = true + has_wiki = true + html_url = (known after apply) + id = (known after apply) + ignore_whitespace_conflicts = false + internal = (known after apply) + internal_tracker = (known after apply) + labels = false + lfs = false + milestones = false + mirror = false + mirror_interval = (known after apply) + mirror_updated = (known after apply) + name = "eats" + open_issues_count = (known after apply) + open_pr_counter = (known after apply) + owner = "ScottyLabs" + parent_id = (known after apply) + permissions = (known after apply) + private = false + release_counter = (known after apply) + size = (known after apply) + ssh_url = (known after apply) + stars_count = (known after apply) + template = false + trust_model = "default" + updated_at = (known after apply) + watchers_count = (known after apply) } # forgejo_team.cmu-eats will be created + resource "forgejo_team" "cmu-eats" { + can_create_org_repo = (known after apply) + description = "CMU Eats" + id = (known after apply) + includes_all_repositories = (known after apply) + name = "cmu-eats" + organization = (known after apply) + organization_id = 3 + permission = "write" + units_map = { + "repo.code" = "write" + "repo.issues" = "write" + "repo.pulls" = "write" } } # forgejo_team.eats_leads will be created + resource "forgejo_team" "eats_leads" { + can_create_org_repo = false + description = "CMU Eats leads" + id = (known after apply) + includes_all_repositories = false + name = "eats-leads" + organization = (known after apply) + organization_id = 3 + permission = "admin" + units_map = { + "repo.actions" = "admin" + "repo.code" = "admin" + "repo.ext_issues" = "admin" + "repo.ext_wiki" = "admin" + "repo.issues" = "admin" + "repo.packages" = "admin" + "repo.projects" = "admin" + "repo.pulls" = "admin" + "repo.releases" = "admin" + "repo.wiki" = "admin" } } # forgejo_team_member.cmu-eats_anish will be created + resource "forgejo_team_member" "cmu-eats_anish" { + team_id = (known after apply) + user = "anish" } # forgejo_team_member.eats_leads_anish will be created + resource "forgejo_team_member" "eats_leads_anish" { + team_id = (known after apply) + user = "anish" } # forgejo_team_repository.cmu-eats_eats will be created + resource "forgejo_team_repository" "cmu-eats_eats" { + owner = "ScottyLabs" + repository = "eats" + team_id = (known after apply) } # forgejo_team_repository.eats_leads_eats will be created + resource "forgejo_team_repository" "eats_leads_eats" { + owner = "ScottyLabs" + repository = "eats" + team_id = (known after apply) } # github_actions_repository_permissions.cmu-eats_eats_actions will be created + resource "github_actions_repository_permissions" "cmu-eats_eats_actions" { + enabled = false + id = (known after apply) + repository = "eats" + sha_pinning_required = (known after apply) } # github_repository.cmu-eats_eats will be created + resource "github_repository" "cmu-eats_eats" { + allow_auto_merge = false + allow_forking = (known after apply) + allow_merge_commit = true + allow_rebase_merge = true + allow_squash_merge = true + archived = false + default_branch = (known after apply) + delete_branch_on_merge = false + description = "READ-ONLY MIRROR" + etag = (known after apply) + fork = (known after apply) + full_name = (known after apply) + git_clone_url = (known after apply) + has_discussions = false + has_issues = false + has_projects = false + has_wiki = false + homepage_url = "https://git.cmu.dev/ScottyLabs/eats" + html_url = (known after apply) + http_clone_url = (known after apply) + id = (known after apply) + ignore_vulnerability_alerts_during_read = false + merge_commit_message = "PR_TITLE" + merge_commit_title = "MERGE_MESSAGE" + name = "eats" + node_id = (known after apply) + primary_language = (known after apply) + private = (known after apply) + repo_id = (known after apply) + source_owner = (known after apply) + source_repo = (known after apply) + squash_merge_commit_message = "COMMIT_MESSAGES" + squash_merge_commit_title = "COMMIT_OR_PR_TITLE" + ssh_clone_url = (known after apply) + svn_url = (known after apply) + topics = (known after apply) + visibility = "public" + vulnerability_alerts = (known after apply) + web_commit_signoff_required = (known after apply) + security_and_analysis (known after apply) } # github_repository.slai_cmugpt_agent will be updated in-place ! resource "github_repository" "slai_cmugpt_agent" { id = "cmugpt-agent" name = "cmugpt-agent" # (39 unchanged attributes hidden) - template { - include_all_branches = false -> null - owner = "ScottyLabs" -> null - repository = "python-template" -> null } # (1 unchanged block hidden) } # github_repository.slai_cmugpt_surface will be updated in-place ! resource "github_repository" "slai_cmugpt_surface" { id = "cmugpt-surface" name = "cmugpt-surface" # (39 unchanged attributes hidden) - template { - include_all_branches = false -> null - owner = "scottylabs-labrador" -> null - repository = "ScottyStack" -> null } # (1 unchanged block hidden) } # github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key will be created + resource "github_repository_deploy_key" "cmu-eats_eats_mirror_deploy_key" { + etag = (known after apply) + id = (known after apply) + key = (known after apply) + read_only = false + repository = "eats" + title = "Forgejo mirror" } # github_repository_ruleset.cbp_bus_sign_ruleset will be updated in-place ! resource "github_repository_ruleset" "cbp_bus_sign_ruleset" { id = "15236550" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665342 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.cbp_dalmatian_ruleset will be updated in-place ! resource "github_repository_ruleset" "cbp_dalmatian_ruleset" { id = "17238915" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665367 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.cbp_discord_verify_ruleset will be updated in-place ! resource "github_repository_ruleset" "cbp_discord_verify_ruleset" { id = "11449154" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665340 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.cbp_groupme_mirror_ruleset will be updated in-place ! resource "github_repository_ruleset" "cbp_groupme_mirror_ruleset" { id = "12503296" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665366 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.cmu-cal_cal_ruleset will be updated in-place ! resource "github_repository_ruleset" "cmu-cal_cal_ruleset" { id = "18919281" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665344 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.cmu-courses_courses_ruleset will be updated in-place ! resource "github_repository_ruleset" "cmu-courses_courses_ruleset" { id = "15855161" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665327 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.cmu-courses_internet_archive_ruleset will be updated in-place ! resource "github_repository_ruleset" "cmu-courses_internet_archive_ruleset" { id = "11988074" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665356 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.cmu-eats_eats_ruleset will be created + resource "github_repository_ruleset" "cmu-eats_eats_ruleset" { + enforcement = "active" + etag = (known after apply) + id = (known after apply) + name = "Default" + node_id = (known after apply) + repository = "eats" + ruleset_id = (known after apply) + target = "branch" + bypass_actors { + actor_id = (known after apply) + actor_type = "DeployKey" + bypass_mode = "always" } + conditions { + ref_name { + exclude = [] + include = [ + "~ALL", ] } } + rules { + creation = true + deletion = true + non_fast_forward = true + update = true + update_allows_fetch_and_merge = false + pull_request { + allowed_merge_methods = (known after apply) + dismiss_stale_reviews_on_push = false + require_code_owner_review = false + require_last_push_approval = false + required_approving_review_count = 1 + required_review_thread_resolution = false } } } # github_repository_ruleset.cmu-housing_housing_ruleset will be updated in-place ! resource "github_repository_ruleset" "cmu-housing_housing_ruleset" { id = "17421604" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665369 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.cmu-maps_maps_ruleset will be updated in-place ! resource "github_repository_ruleset" "cmu-maps_maps_ruleset" { id = "18864961" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665325 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.devops_collie_ruleset will be updated in-place ! resource "github_repository_ruleset" "devops_collie_ruleset" { id = "18865204" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665332 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.devops_documentation_ruleset will be updated in-place ! resource "github_repository_ruleset" "devops_documentation_ruleset" { id = "17079368" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665362 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.devops_governance_ruleset will be updated in-place ! resource "github_repository_ruleset" "devops_governance_ruleset" { id = "14488792" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665346 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.devops_infrastructure_ruleset will be updated in-place ! resource "github_repository_ruleset" "devops_infrastructure_ruleset" { id = "11275118" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665343 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.devops_kennel_ruleset will be updated in-place ! resource "github_repository_ruleset" "devops_kennel_ruleset" { id = "13193453" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665328 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.devops_link_shortener_ruleset will be updated in-place ! resource "github_repository_ruleset" "devops_link_shortener_ruleset" { id = "17911647" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665345 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset will be updated in-place ! resource "github_repository_ruleset" "lost-and-found_lost_and_found_v2_ruleset" { id = "18920618" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665337 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.quest_quest_ruleset will be updated in-place ! resource "github_repository_ruleset" "quest_quest_ruleset" { id = "15750052" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665323 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.slai_cmugpt_agent_ruleset will be updated in-place ! resource "github_repository_ruleset" "slai_cmugpt_agent_ruleset" { id = "16476661" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665360 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.slai_cmugpt_sms_surface_ruleset will be updated in-place ! resource "github_repository_ruleset" "slai_cmugpt_sms_surface_ruleset" { id = "17653622" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665365 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.slai_cmugpt_surface_ruleset will be updated in-place ! resource "github_repository_ruleset" "slai_cmugpt_surface_ruleset" { id = "16476662" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665348 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.slai_mcp_server_ruleset will be updated in-place ! resource "github_repository_ruleset" "slai_mcp_server_ruleset" { id = "17082154" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665358 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.tartan-vote_tartan_vote_ruleset will be updated in-place ! resource "github_repository_ruleset" "tartan-vote_tartan_vote_ruleset" { id = "16848966" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665350 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.terrier_terrier_ruleset will be updated in-place ! resource "github_repository_ruleset" "terrier_terrier_ruleset" { id = "12671003" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665339 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.uia_components_ruleset will be updated in-place ! resource "github_repository_ruleset" "uia_components_ruleset" { id = "16366085" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665353 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_repository_ruleset.uia_scottylabs_org_ruleset will be updated in-place ! resource "github_repository_ruleset" "uia_scottylabs_org_ruleset" { id = "18112214" name = "Default" # (6 unchanged attributes hidden) ! bypass_actors { ! actor_id = 0 -> 159665361 # (2 unchanged attributes hidden) } # (2 unchanged blocks hidden) } # github_team.cmu-eats will be created + resource "github_team" "cmu-eats" { + create_default_maintainer = false + etag = (known after apply) + id = (known after apply) + members_count = (known after apply) + name = "CMU Eats" + node_id = (known after apply) + notification_setting = "notifications_enabled" + parent_team_read_id = (known after apply) + parent_team_read_slug = (known after apply) + privacy = "closed" + slug = (known after apply) } # github_team_membership.cmu-eats_anish will be created + resource "github_team_membership" "cmu-eats_anish" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = (known after apply) + username = "ap-1" } # github_team_membership.quest_brightz will be created + resource "github_team_membership" "quest_brightz" { + etag = (known after apply) + id = (known after apply) + role = "member" + team_id = "17330197" + username = "bright-zheng" } # restapi_object.cmu-eats_eats_kennel_webhook will be created + resource "restapi_object" "cmu-eats_eats_kennel_webhook" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = (sensitive value) + id = (known after apply) + id_attribute = "id" + path = "/api/v1/repos/ScottyLabs/eats/hooks" + update_method = "PATCH" } # restapi_object.cmu-eats_eats_push_mirror will be created + resource "restapi_object" "cmu-eats_eats_push_mirror" { + api_data = (known after apply) + api_response = (known after apply) + create_response = (known after apply) + data = jsonencode( { + interval = "8h0m0s" + remote_address = "git@github.com:ScottyLabs/eats.git" + sync_on_commit = true + use_ssh = true } ) + id = (known after apply) + id_attribute = "remote_name" + path = "/api/v1/repos/ScottyLabs/eats/push_mirrors" } Plan: 16 to add, 27 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 16 to add, 27 to change, 0 to destroy. --- ### Plan Summary 9 projects, 9 with changes, 0 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Author
Owner

atlantis apply

atlantis apply

Ran Apply for 9 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: discord dir: tofu/discord workspace: default
  3. project: openbao dir: tofu/openbao workspace: default
  4. project: posthog dir: tofu/posthog workspace: default
  5. project: google_play dir: tofu/google_play workspace: default
  6. project: sentry dir: tofu/sentry workspace: default
  7. project: keycloak dir: tofu/keycloak workspace: default
  8. project: matrix_bridges dir: tofu/matrix_bridges workspace: default
  9. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.48s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_cmu-eats_ch0_anish: Creating...
null_resource.slack_cmu-eats_ch0_anish: Provisioning with 'local-exec'...
null_resource.slack_cmu-eats_ch0_anish (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C043UV30DHA --user U07FPJKFB5E"]
null_resource.slack_cmu-eats_ch0_anish: Creation complete after 0s [id=880475801292066511]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

2. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.14s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
discord_role.cmu-eats: Creating...
discord_role.cmu-eats: Creation complete after 1s [id=1541552714063024148]
discord_channel_permission.cmu-eats_ch0: Creating...
discord_member_roles.anish: Modifying... [id=1353118209024331948:315249559473225729]
discord_channel_permission.cmu-eats_ch0: Creation complete after 0s [id=1541336561105240134:1541552714063024148:role]
discord_member_roles.anish: Modifications complete after 0s [id=1353118209024331948:315249559473225729]

Apply complete! Resources: 2 added, 1 changed, 0 destroyed.

3. project: openbao dir: tofu/openbao workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.26s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
vault_policy.eats_prod: Creating...
vault_identity_group.eats_members: Creating...
vault_policy.eats_dev: Creating...
vault_identity_group.eats_admins: Creating...
vault_identity_group.eats_admins: Creation complete after 0s [id=05959815-c896-9703-32eb-7ba84604303c]
vault_identity_group.eats_members: Creation complete after 0s [id=7bece731-c279-57ce-e2cc-cba57e4a050d]
vault_identity_group_alias.eats_members: Creating...
vault_identity_group_alias.eats_admins: Creating...
vault_policy.eats_dev: Creation complete after 0s [id=eats-dev]
vault_policy.eats_prod: Creation complete after 0s [id=eats-prod]
vault_identity_group_alias.eats_members: Creation complete after 0s [id=d5a06eaf-a547-81e9-3960-362b95c6b7cd]
vault_identity_group_alias.eats_admins: Creation complete after 0s [id=e6706408-5947-6868-b119-563fc11b255c]

Apply complete! Resources: 6 added, 0 changed, 0 destroyed.

4. project: posthog dir: tofu/posthog workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.50s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
posthog_project.cmu-eats_eats: Creating...
vault_kv_secret_v2.cmu-eats_eats_posthog_host: Creating...
vault_kv_secret_v2.cmu-eats_eats_posthog_host: Creation complete after 0s [id=secret/data/secretspec/eats/prod/POSTHOG_HOST]
posthog_project.cmu-eats_eats: Creation complete after 1s [name=eats]
vault_kv_secret_v2.cmu-eats_eats_posthog_key: Creating...
vault_kv_secret_v2.cmu-eats_eats_posthog_key: Creation complete after 0s [id=secret/data/secretspec/eats/prod/POSTHOG_KEY]

Apply complete! Resources: 3 added, 0 changed, 0 destroyed.

5. project: google_play dir: tofu/google_play workspace: default

Apply Error

Show Output
running 'sh -c' '/nix/store/9dj5sdaq6vksi96ibn9vpyzjcbhf26jv-opentofu-1.12.5/bin/tofu apply -input=false "/var/lib/atlantis/repos/ScottyLabs/governance/178/default/tofu/google_play/google_play-default.tfplan"' in '/var/lib/atlantis/repos/ScottyLabs/governance/178/default/tofu/google_play': exit status 1
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.17s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
googleplay_user.brightz: Creating...
googleplay_user.kothurvidur: Creating...
googleplay_user.anish: Creating...
googleplay_user.dkou: Creating...
googleplay_user.xboxbedrock: Creating...
googleplay_user.kechezon: Creating...
googleplay_user.kothurvidur: Creation complete after 1s [id=vkothur@andrew.cmu.edu]
googleplay_app_iam.terrier_kothurvidur: Creating...
googleplay_user.dkou: Creation complete after 1s [id=dkou@andrew.cmu.edu]
googleplay_app_iam.terrier_dkou: Creating...
googleplay_user.xboxbedrock: Creation complete after 1s [id=spola2@andrew.cmu.edu]
googleplay_app_iam.quest_xboxbedrock: Creating...
googleplay_user.brightz: Creation complete after 1s [id=brightz@andrew.cmu.edu]
googleplay_app_iam.quest_brightz: Creating...
googleplay_user.anish: Creation complete after 1s [id=apallati@andrew.cmu.edu]
googleplay_user.kechezon: Creation complete after 1s [id=kechezon@andrew.cmu.edu]
googleplay_app_iam.quest_kechezon: Creating...
╷
│ Error: Provider produced inconsistent result after apply
│ 
│ When applying changes to googleplay_app_iam.quest_brightz, provider
│ "provider[\"registry.terraform.io/oliver-binns/googleplay\"]" produced an
│ unexpected new value: .app_id: was cty.StringVal("4972435805279727996"),
│ but now cty.StringVal("quest.cmu.twa").
│ 
│ This is a bug in the provider, which should be reported in the provider's
│ own issue tracker.
╵
╷
│ Error: Provider produced inconsistent result after apply
│ 
│ When applying changes to googleplay_app_iam.terrier_kothurvidur, provider
│ "provider[\"registry.terraform.io/oliver-binns/googleplay\"]" produced an
│ unexpected new value: .app_id: was cty.StringVal("4975152612825535266"),
│ but now cty.StringVal("build.terrier.app").
│ 
│ This is a bug in the provider, which should be reported in the provider's
│ own issue tracker.
╵
╷
│ Error: Provider produced inconsistent result after apply
│ 
│ When applying changes to googleplay_app_iam.quest_xboxbedrock, provider
│ "provider[\"registry.terraform.io/oliver-binns/googleplay\"]" produced an
│ unexpected new value: .app_id: was cty.StringVal("4972435805279727996"),
│ but now cty.StringVal("quest.cmu.twa").
│ 
│ This is a bug in the provider, which should be reported in the provider's
│ own issue tracker.
╵
╷
│ Error: Provider produced inconsistent result after apply
│ 
│ When applying changes to googleplay_app_iam.terrier_dkou, provider
│ "provider[\"registry.terraform.io/oliver-binns/googleplay\"]" produced an
│ unexpected new value: .app_id: was cty.StringVal("4975152612825535266"),
│ but now cty.StringVal("build.terrier.app").
│ 
│ This is a bug in the provider, which should be reported in the provider's
│ own issue tracker.
╵
╷
│ Error: Provider produced inconsistent result after apply
│ 
│ When applying changes to googleplay_app_iam.quest_kechezon, provider
│ "provider[\"registry.terraform.io/oliver-binns/googleplay\"]" produced an
│ unexpected new value: .app_id: was cty.StringVal("4972435805279727996"),
│ but now cty.StringVal("quest.cmu.twa").
│ 
│ This is a bug in the provider, which should be reported in the provider's
│ own issue tracker.
╵


6. project: sentry dir: tofu/sentry workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.37s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
sentry_team.cmu-eats: Creating...
sentry_team.cmu-eats: Creation complete after 1s [id=cmu-eats]
sentry_project.cmu-eats_eats: Creating...
sentry_project.cmu-eats_eats: Creation complete after 1s [id=eats]
sentry_key.cmu-eats_eats: Creating...
sentry_key.cmu-eats_eats: Creation complete after 1s [id=46a2c57055681e48cc1891f4168268af]
vault_kv_secret_v2.cmu-eats_eats_sentry_dsn: Creating...
vault_kv_secret_v2.cmu-eats_eats_sentry_dsn: Creation complete after 0s [id=secret/data/secretspec/eats/prod/SENTRY_DSN]

Apply complete! Resources: 4 added, 0 changed, 0 destroyed.

7. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.36s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
vault_kv_secret_v2.eats_staging_oauth_relay_url: Creating...
vault_kv_secret_v2.eats_prod_oauth_relay_url: Creating...
vault_kv_secret_v2.eats_preview_keycloak_url: Creating...
vault_kv_secret_v2.eats_dev_keycloak_url: Creating...
vault_kv_secret_v2.eats_prod_project_group: Creating...
vault_kv_secret_v2.eats_dev_oauth_relay_url: Creating...
vault_kv_secret_v2.eats_staging_keycloak_realm: Creating...
vault_kv_secret_v2.eats_staging_project_admin_group: Creating...
vault_kv_secret_v2.eats_staging_project_group: Creating...
vault_kv_secret_v2.eats_prod_keycloak_url: Creating...
vault_kv_secret_v2.eats_dev_keycloak_url: Creation complete after 0s [id=secret/data/secretspec/eats/dev/KEYCLOAK_URL]
vault_kv_secret_v2.eats_staging_oauth_relay_url: Creation complete after 0s [id=secret/data/secretspec/eats/staging/OAUTH_RELAY_URL]
vault_kv_secret_v2.eats_dev_oauth_relay_url: Creation complete after 0s [id=secret/data/secretspec/eats/dev/OAUTH_RELAY_URL]
vault_kv_secret_v2.eats_prod_oauth_relay_url: Creation complete after 0s [id=secret/data/secretspec/eats/prod/OAUTH_RELAY_URL]
vault_kv_secret_v2.eats_staging_project_admin_group: Creation complete after 0s [id=secret/data/secretspec/eats/staging/PROJECT_ADMIN_GROUP]
vault_kv_secret_v2.eats_preview_oauth_relay_url: Creating...
vault_kv_secret_v2.eats_prod_keycloak_realm: Creating...
vault_kv_secret_v2.eats_preview_project_group: Creating...
vault_kv_secret_v2.eats_staging_keycloak_url: Creating...
vault_kv_secret_v2.eats_dev_project_admin_group: Creating...
vault_kv_secret_v2.eats_staging_keycloak_realm: Creation complete after 0s [id=secret/data/secretspec/eats/staging/KEYCLOAK_REALM]
vault_kv_secret_v2.eats_preview_keycloak_url: Creation complete after 0s [id=secret/data/secretspec/eats/preview/KEYCLOAK_URL]
vault_kv_secret_v2.eats_dev_project_group: Creating...
vault_kv_secret_v2.eats_prod_project_group: Creation complete after 0s [id=secret/data/secretspec/eats/prod/PROJECT_GROUP]
vault_kv_secret_v2.eats_dev_keycloak_realm: Creating...
vault_kv_secret_v2.eats_staging_project_group: Creation complete after 0s [id=secret/data/secretspec/eats/staging/PROJECT_GROUP]
vault_kv_secret_v2.eats_prod_keycloak_url: Creation complete after 0s [id=secret/data/secretspec/eats/prod/KEYCLOAK_URL]
vault_kv_secret_v2.eats_preview_project_admin_group: Creating...
vault_kv_secret_v2.eats_prod_project_admin_group: Creating...
vault_kv_secret_v2.eats_preview_keycloak_realm: Creating...
vault_kv_secret_v2.eats_prod_keycloak_realm: Creation complete after 1s [id=secret/data/secretspec/eats/prod/KEYCLOAK_REALM]
vault_kv_secret_v2.eats_staging_keycloak_url: Creation complete after 1s [id=secret/data/secretspec/eats/staging/KEYCLOAK_URL]
vault_kv_secret_v2.eats_preview_oauth_relay_url: Creation complete after 1s [id=secret/data/secretspec/eats/preview/OAUTH_RELAY_URL]
vault_kv_secret_v2.eats_preview_project_admin_group: Creation complete after 1s [id=secret/data/secretspec/eats/preview/PROJECT_ADMIN_GROUP]
vault_kv_secret_v2.eats_preview_project_group: Creation complete after 1s [id=secret/data/secretspec/eats/preview/PROJECT_GROUP]
vault_kv_secret_v2.eats_dev_keycloak_realm: Creation complete after 1s [id=secret/data/secretspec/eats/dev/KEYCLOAK_REALM]
keycloak_openid_client.eats: Creating...
vault_kv_secret_v2.eats_preview_keycloak_realm: Creation complete after 1s [id=secret/data/secretspec/eats/preview/KEYCLOAK_REALM]
vault_kv_secret_v2.eats_prod_project_admin_group: Creation complete after 1s [id=secret/data/secretspec/eats/prod/PROJECT_ADMIN_GROUP]
keycloak_openid_client.eats_staging: Creating...
keycloak_group.project_cmu-eats_eats: Creating...
keycloak_openid_client.eats_dev: Creating...
vault_kv_secret_v2.eats_dev_project_group: Creation complete after 1s [id=secret/data/secretspec/eats/dev/PROJECT_GROUP]
vault_kv_secret_v2.eats_dev_project_admin_group: Creation complete after 1s [id=secret/data/secretspec/eats/dev/PROJECT_ADMIN_GROUP]
keycloak_group.project_cmu-eats_eats: Creation complete after 0s [id=41a1f091-ff21-421d-b98a-085de9cd77ba]
keycloak_group.project_cmu-eats_eats_admins: Creating...
keycloak_group_memberships.project_cmu-eats_eats: Creating...
keycloak_openid_client.eats_staging: Creation complete after 0s [id=a17ef8d4-bd16-47d0-b10d-0efb1e800236]
keycloak_openid_group_membership_protocol_mapper.eats_staging: Creating...
vault_kv_secret_v2.eats_staging_oidc_client_id: Creating...
vault_kv_secret_v2.eats_staging_oidc_client_secret: Creating...
vault_kv_secret_v2.eats_preview_oidc_client_id: Creating...
vault_kv_secret_v2.eats_preview_oidc_client_secret: Creating...
keycloak_openid_client.eats_dev: Creation complete after 0s [id=696328b5-e9b0-4331-8442-5ffbe79aba7f]
keycloak_openid_group_membership_protocol_mapper.eats_dev: Creating...
vault_kv_secret_v2.eats_dev_oidc_client_secret: Creating...
keycloak_openid_client.eats: Creation complete after 0s [id=fe3cf80b-e521-42a1-aff4-164cca377dd1]
vault_kv_secret_v2.eats_dev_oidc_client_id: Creating...
keycloak_group.project_cmu-eats_eats_admins: Creation complete after 0s [id=1a583ab1-469d-4eb4-a5a8-f2fd143ba829]
vault_kv_secret_v2.eats_staging_oidc_client_secret: Creation complete after 0s [id=secret/data/secretspec/eats/staging/OIDC_CLIENT_SECRET]
vault_kv_secret_v2.eats_prod_oidc_client_id: Creating...
keycloak_openid_group_membership_protocol_mapper.eats: Creating...
vault_kv_secret_v2.eats_staging_oidc_client_id: Creation complete after 0s [id=secret/data/secretspec/eats/staging/OIDC_CLIENT_ID]
vault_kv_secret_v2.eats_prod_oidc_client_secret: Creating...
vault_kv_secret_v2.eats_preview_oidc_client_secret: Creation complete after 0s [id=secret/data/secretspec/eats/preview/OIDC_CLIENT_SECRET]
vault_kv_secret_v2.eats_preview_oidc_client_id: Creation complete after 0s [id=secret/data/secretspec/eats/preview/OIDC_CLIENT_ID]
keycloak_group_memberships.project_cmu-eats_eats_admins: Creating...
keycloak_openid_group_membership_protocol_mapper.eats_staging: Creation complete after 0s [id=1c2c928c-1af7-40ea-a10e-b654a0f35e4e]
vault_kv_secret_v2.eats_dev_oidc_client_secret: Creation complete after 0s [id=secret/data/secretspec/eats/dev/OIDC_CLIENT_SECRET]
keycloak_openid_group_membership_protocol_mapper.eats_dev: Creation complete after 0s [id=c537d1d2-f898-4a7a-ac30-d313657cdcb2]
vault_kv_secret_v2.eats_dev_oidc_client_id: Creation complete after 0s [id=secret/data/secretspec/eats/dev/OIDC_CLIENT_ID]
vault_kv_secret_v2.eats_prod_oidc_client_id: Creation complete after 0s [id=secret/data/secretspec/eats/prod/OIDC_CLIENT_ID]
vault_kv_secret_v2.eats_prod_oidc_client_secret: Creation complete after 0s [id=secret/data/secretspec/eats/prod/OIDC_CLIENT_SECRET]
keycloak_openid_group_membership_protocol_mapper.eats: Creation complete after 0s [id=d461ed68-2993-459b-9f87-f22d761eb29e]
keycloak_group_memberships.project_cmu-eats_eats: Creation complete after 3s [id=scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba]
keycloak_group_memberships.project_cmu-eats_eats_admins: Creation complete after 3s [id=scottylabs/group-memberships/1a583ab1-469d-4eb4-a5a8-f2fd143ba829]

Apply complete! Resources: 38 added, 0 changed, 0 destroyed.

8. project: matrix_bridges dir: tofu/matrix_bridges workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.32s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
null_resource.slack_bridge_login_cmu_eats: Creating...
null_resource.slack_bridge_login_cmu_eats: Provisioning with 'local-exec'...
null_resource.slack_bridge_login_cmu_eats (local-exec): Executing: ["/bin/sh" "-c" "governance slack-join --channel C043UV30DHA"]
null_resource.slack_bridge_login_cmu_eats: Creation complete after 0s [id=698518873962630849]
synapse_mautrix_slack_link.cmu_eats: Creating...
synapse_mautrix_slack_link.cmu_eats: Still creating... [10s elapsed]
synapse_mautrix_slack_link.cmu_eats: Creation complete after 17s [id=1541336561105240134]

Apply complete! Resources: 2 added, 0 changed, 0 destroyed.

9. project: git dir: tofu/git workspace: default

Apply Error

Show Output
running 'sh -c' '/nix/store/9dj5sdaq6vksi96ibn9vpyzjcbhf26jv-opentofu-1.12.5/bin/tofu apply -input=false "/var/lib/atlantis/repos/ScottyLabs/governance/178/default/tofu/git/git-default.tfplan"' in '/var/lib/atlantis/repos/ScottyLabs/governance/178/default/tofu/git': exit status 1
    Finished `release` profile [optimized] target(s) in 0.13s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
github_team_membership.quest_brightz: Creating...
github_team.cmu-eats: Creating...
github_repository.cmu-eats_eats: Creating...
github_repository.slai_cmugpt_agent: Modifying... [id=cmugpt-agent]
github_repository.slai_cmugpt_surface: Modifying... [id=cmugpt-surface]
github_repository_ruleset.devops_kennel_ruleset: Modifying... [id=13193453]
github_repository_ruleset.cbp_bus_sign_ruleset: Modifying... [id=15236550]
github_repository_ruleset.cbp_discord_verify_ruleset: Modifying... [id=11449154]
github_repository_ruleset.devops_infrastructure_ruleset: Modifying... [id=11275118]
github_repository_ruleset.uia_scottylabs_org_ruleset: Modifying... [id=18112214]
forgejo_repository.cmu-eats_eats: Creating...
forgejo_repository.cmu-eats_eats: Creation complete after 1s [name=eats]
forgejo_team.eats_leads: Creating...
forgejo_team.eats_leads: Creation complete after 0s [name=eats-leads]
forgejo_team.cmu-eats: Creating...
forgejo_team.cmu-eats: Creation complete after 0s [name=cmu-eats]
restapi_object.cmu-eats_eats_push_mirror: Creating...
restapi_object.cmu-eats_eats_push_mirror: Creation complete after 0s [id=remote_mirror_KOBYPLZTKSZ]
restapi_object.cmu-eats_eats_kennel_webhook: Creating...
restapi_object.cmu-eats_eats_kennel_webhook: Creation complete after 0s [id=30]
github_repository_ruleset.quest_quest_ruleset: Modifying... [id=15750052]
github_team_membership.quest_brightz: Creation complete after 9s [id=17330197:bright-zheng]
github_repository_ruleset.devops_governance_ruleset: Modifying... [id=14488792]
github_repository.cmu-eats_eats: Still creating... [10s elapsed]
github_repository.slai_cmugpt_agent: Still modifying... [id=cmugpt-agent, 10s elapsed]
github_repository.slai_cmugpt_surface: Still modifying... [id=cmugpt-surface, 10s elapsed]
github_repository_ruleset.devops_kennel_ruleset: Still modifying... [id=13193453, 10s elapsed]
github_repository_ruleset.cbp_bus_sign_ruleset: Still modifying... [id=15236550, 10s elapsed]
github_repository_ruleset.cbp_discord_verify_ruleset: Still modifying... [id=11449154, 10s elapsed]
github_repository_ruleset.devops_infrastructure_ruleset: Still modifying... [id=11275118, 10s elapsed]
github_repository_ruleset.uia_scottylabs_org_ruleset: Still modifying... [id=18112214, 10s elapsed]
github_repository.slai_cmugpt_agent: Modifications complete after 11s [id=cmugpt-agent]
github_repository_ruleset.devops_collie_ruleset: Modifying... [id=18865204]
github_repository.slai_cmugpt_surface: Modifications complete after 12s [id=cmugpt-surface]
github_repository_ruleset.cmu-maps_maps_ruleset: Modifying... [id=18864961]
github_repository_ruleset.quest_quest_ruleset: Still modifying... [id=15750052, 10s elapsed]
github_repository_ruleset.devops_kennel_ruleset: Modifications complete after 12s [id=13193453]
github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Modifying... [id=16848966]
github_repository_ruleset.cbp_discord_verify_ruleset: Modifications complete after 14s [id=11449154]
github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Modifying... [id=17653622]
github_repository_ruleset.cbp_bus_sign_ruleset: Modifications complete after 15s [id=15236550]
github_repository_ruleset.cbp_dalmatian_ruleset: Modifying... [id=17238915]
github_repository_ruleset.devops_infrastructure_ruleset: Modifications complete after 16s [id=11275118]
github_repository_ruleset.devops_documentation_ruleset: Modifying... [id=17079368]
github_repository_ruleset.uia_scottylabs_org_ruleset: Modifications complete after 18s [id=18112214]
github_repository_ruleset.cmu-housing_housing_ruleset: Modifying... [id=17421604]
github_repository_ruleset.devops_governance_ruleset: Still modifying... [id=14488792, 10s elapsed]
github_repository_ruleset.quest_quest_ruleset: Modifications complete after 17s [id=15750052]
github_repository_ruleset.cmu-courses_internet_archive_ruleset: Modifying... [id=11988074]
github_repository.cmu-eats_eats: Still creating... [20s elapsed]
github_repository_ruleset.devops_collie_ruleset: Still modifying... [id=18865204, 10s elapsed]
github_repository_ruleset.cmu-maps_maps_ruleset: Still modifying... [id=18864961, 10s elapsed]
github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Still modifying... [id=16848966, 10s elapsed]
github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Still modifying... [id=17653622, 10s elapsed]
github_repository_ruleset.devops_governance_ruleset: Modifications complete after 15s [id=14488792]
github_repository_ruleset.cmu-courses_courses_ruleset: Modifying... [id=15855161]
github_repository_ruleset.cbp_dalmatian_ruleset: Still modifying... [id=17238915, 10s elapsed]
github_repository_ruleset.devops_collie_ruleset: Modifications complete after 15s [id=18865204]
forgejo_team_repository.eats_leads_eats: Creating...
forgejo_team_repository.eats_leads_eats: Creation complete after 0s
forgejo_team_member.eats_leads_anish: Creating...
forgejo_team_member.eats_leads_anish: Creation complete after 0s
github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Modifying... [id=18920618]
github_repository_ruleset.devops_documentation_ruleset: Still modifying... [id=17079368, 10s elapsed]
github_repository_ruleset.cmu-maps_maps_ruleset: Modifications complete after 15s [id=18864961]
github_repository_ruleset.slai_mcp_server_ruleset: Modifying... [id=17082154]
github_repository_ruleset.cmu-housing_housing_ruleset: Still modifying... [id=17421604, 10s elapsed]
github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Modifications complete after 17s [id=16848966]
github_repository_ruleset.terrier_terrier_ruleset: Modifying... [id=12671003]
github_repository_ruleset.cmu-courses_internet_archive_ruleset: Still modifying... [id=11988074, 10s elapsed]
github_repository.cmu-eats_eats: Still creating... [30s elapsed]
github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Modifications complete after 16s [id=17653622]
forgejo_team_member.cmu-eats_anish: Creating...
forgejo_team_member.cmu-eats_anish: Creation complete after 0s
forgejo_team_repository.cmu-eats_eats: Creating...
forgejo_team_repository.cmu-eats_eats: Creation complete after 0s
github_repository_ruleset.cbp_groupme_mirror_ruleset: Modifying... [id=12503296]
github_repository_ruleset.cbp_dalmatian_ruleset: Modifications complete after 17s [id=17238915]
github_repository_ruleset.devops_link_shortener_ruleset: Modifying... [id=17911647]
github_repository_ruleset.devops_documentation_ruleset: Modifications complete after 17s [id=17079368]
github_repository_ruleset.uia_components_ruleset: Modifying... [id=16366085]
github_repository_ruleset.cmu-courses_courses_ruleset: Still modifying... [id=15855161, 10s elapsed]
github_repository_ruleset.cmu-housing_housing_ruleset: Modifications complete after 17s [id=17421604]
github_repository_ruleset.cmu-cal_cal_ruleset: Modifying... [id=18919281]
github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Still modifying... [id=18920618, 10s elapsed]
github_repository_ruleset.cmu-courses_internet_archive_ruleset: Modifications complete after 17s [id=11988074]
github_repository_ruleset.slai_cmugpt_agent_ruleset: Modifying... [id=16476661]
github_repository_ruleset.slai_mcp_server_ruleset: Still modifying... [id=17082154, 10s elapsed]
github_repository.cmu-eats_eats: Creation complete after 37s [id=eats]
github_repository_ruleset.slai_cmugpt_surface_ruleset: Modifying... [id=16476662]
github_repository_ruleset.terrier_terrier_ruleset: Still modifying... [id=12671003, 10s elapsed]
github_repository_ruleset.cmu-courses_courses_ruleset: Modifications complete after 16s [id=15855161]
github_actions_repository_permissions.cmu-eats_eats_actions: Creating...
github_repository_ruleset.cbp_groupme_mirror_ruleset: Still modifying... [id=12503296, 10s elapsed]
github_repository_ruleset.devops_link_shortener_ruleset: Still modifying... [id=17911647, 10s elapsed]
github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Modifications complete after 16s [id=18920618]
github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Creating...
github_repository_ruleset.slai_mcp_server_ruleset: Modifications complete after 16s [id=17082154]
github_repository_ruleset.uia_components_ruleset: Still modifying... [id=16366085, 10s elapsed]
github_repository_ruleset.terrier_terrier_ruleset: Modifications complete after 15s [id=12671003]
github_repository_ruleset.cmu-cal_cal_ruleset: Still modifying... [id=18919281, 10s elapsed]
github_repository_ruleset.cbp_groupme_mirror_ruleset: Modifications complete after 16s [id=12503296]
github_repository_ruleset.slai_cmugpt_agent_ruleset: Still modifying... [id=16476661, 10s elapsed]
github_repository_ruleset.slai_cmugpt_surface_ruleset: Still modifying... [id=16476662, 10s elapsed]
github_repository_ruleset.devops_link_shortener_ruleset: Modifications complete after 15s [id=17911647]
github_repository_ruleset.uia_components_ruleset: Modifications complete after 16s [id=16366085]
github_actions_repository_permissions.cmu-eats_eats_actions: Still creating... [10s elapsed]
github_repository_ruleset.cmu-cal_cal_ruleset: Modifications complete after 15s [id=18919281]
github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Still creating... [10s elapsed]
github_repository_ruleset.slai_cmugpt_agent_ruleset: Modifications complete after 17s [id=16476661]
github_repository_ruleset.slai_cmugpt_surface_ruleset: Modifications complete after 17s [id=16476662]
github_actions_repository_permissions.cmu-eats_eats_actions: Creation complete after 17s [id=eats]
github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Creation complete after 15s [id=eats:161201842]
github_repository_ruleset.cmu-eats_eats_ruleset: Creating...
github_repository_ruleset.cmu-eats_eats_ruleset: Creation complete after 0s [id=21330218]
╷
│ Error: POST https://api.github.com/orgs/ScottyLabs/teams: 422 Validation Failed [{Resource:Team Field:data Code:unprocessable Message:Name must be unique for this org}]
│ 
│   with github_team.cmu-eats,
│   on github_teams.tf.json line 19, in resource.github_team.cmu-eats:
│   19:       },
│ 
╵


Apply Summary

9 projects, 7 successful, 0 failed, 2 errored

Ran Apply for 9 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `openbao` dir: `tofu/openbao` workspace: `default` 1. project: `posthog` dir: `tofu/posthog` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `sentry` dir: `tofu/sentry` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.48s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_cmu-eats_ch0_anish: Creating... null_resource.slack_cmu-eats_ch0_anish: Provisioning with 'local-exec'... null_resource.slack_cmu-eats_ch0_anish (local-exec): Executing: ["/bin/sh" "-c" "governance slack-invite --channel C043UV30DHA --user U07FPJKFB5E"] null_resource.slack_cmu-eats_ch0_anish: Creation complete after 0s [id=880475801292066511] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.14s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ discord_role.cmu-eats: Creating... discord_role.cmu-eats: Creation complete after 1s [id=1541552714063024148] discord_channel_permission.cmu-eats_ch0: Creating... discord_member_roles.anish: Modifying... [id=1353118209024331948:315249559473225729] discord_channel_permission.cmu-eats_ch0: Creation complete after 0s [id=1541336561105240134:1541552714063024148:role] discord_member_roles.anish: Modifications complete after 0s [id=1353118209024331948:315249559473225729] Apply complete! Resources: 2 added, 1 changed, 0 destroyed. ``` </details> --- ### 3. project: `openbao` dir: `tofu/openbao` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.26s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ vault_policy.eats_prod: Creating... vault_identity_group.eats_members: Creating... vault_policy.eats_dev: Creating... vault_identity_group.eats_admins: Creating... vault_identity_group.eats_admins: Creation complete after 0s [id=05959815-c896-9703-32eb-7ba84604303c] vault_identity_group.eats_members: Creation complete after 0s [id=7bece731-c279-57ce-e2cc-cba57e4a050d] vault_identity_group_alias.eats_members: Creating... vault_identity_group_alias.eats_admins: Creating... vault_policy.eats_dev: Creation complete after 0s [id=eats-dev] vault_policy.eats_prod: Creation complete after 0s [id=eats-prod] vault_identity_group_alias.eats_members: Creation complete after 0s [id=d5a06eaf-a547-81e9-3960-362b95c6b7cd] vault_identity_group_alias.eats_admins: Creation complete after 0s [id=e6706408-5947-6868-b119-563fc11b255c] Apply complete! Resources: 6 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `posthog` dir: `tofu/posthog` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.50s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ posthog_project.cmu-eats_eats: Creating... vault_kv_secret_v2.cmu-eats_eats_posthog_host: Creating... vault_kv_secret_v2.cmu-eats_eats_posthog_host: Creation complete after 0s [id=secret/data/secretspec/eats/prod/POSTHOG_HOST] posthog_project.cmu-eats_eats: Creation complete after 1s [name=eats] vault_kv_secret_v2.cmu-eats_eats_posthog_key: Creating... vault_kv_secret_v2.cmu-eats_eats_posthog_key: Creation complete after 0s [id=secret/data/secretspec/eats/prod/POSTHOG_KEY] Apply complete! Resources: 3 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `google_play` dir: `tofu/google_play` workspace: `default` **Apply Error** <details><summary>Show Output</summary> ``` running 'sh -c' '/nix/store/9dj5sdaq6vksi96ibn9vpyzjcbhf26jv-opentofu-1.12.5/bin/tofu apply -input=false "/var/lib/atlantis/repos/ScottyLabs/governance/178/default/tofu/google_play/google_play-default.tfplan"' in '/var/lib/atlantis/repos/ScottyLabs/governance/178/default/tofu/google_play': exit status 1 Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.17s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ googleplay_user.brightz: Creating... googleplay_user.kothurvidur: Creating... googleplay_user.anish: Creating... googleplay_user.dkou: Creating... googleplay_user.xboxbedrock: Creating... googleplay_user.kechezon: Creating... googleplay_user.kothurvidur: Creation complete after 1s [id=vkothur@andrew.cmu.edu] googleplay_app_iam.terrier_kothurvidur: Creating... googleplay_user.dkou: Creation complete after 1s [id=dkou@andrew.cmu.edu] googleplay_app_iam.terrier_dkou: Creating... googleplay_user.xboxbedrock: Creation complete after 1s [id=spola2@andrew.cmu.edu] googleplay_app_iam.quest_xboxbedrock: Creating... googleplay_user.brightz: Creation complete after 1s [id=brightz@andrew.cmu.edu] googleplay_app_iam.quest_brightz: Creating... googleplay_user.anish: Creation complete after 1s [id=apallati@andrew.cmu.edu] googleplay_user.kechezon: Creation complete after 1s [id=kechezon@andrew.cmu.edu] googleplay_app_iam.quest_kechezon: Creating... ╷ │ Error: Provider produced inconsistent result after apply │ │ When applying changes to googleplay_app_iam.quest_brightz, provider │ "provider[\"registry.terraform.io/oliver-binns/googleplay\"]" produced an │ unexpected new value: .app_id: was cty.StringVal("4972435805279727996"), │ but now cty.StringVal("quest.cmu.twa"). │ │ This is a bug in the provider, which should be reported in the provider's │ own issue tracker. ╵ ╷ │ Error: Provider produced inconsistent result after apply │ │ When applying changes to googleplay_app_iam.terrier_kothurvidur, provider │ "provider[\"registry.terraform.io/oliver-binns/googleplay\"]" produced an │ unexpected new value: .app_id: was cty.StringVal("4975152612825535266"), │ but now cty.StringVal("build.terrier.app"). │ │ This is a bug in the provider, which should be reported in the provider's │ own issue tracker. ╵ ╷ │ Error: Provider produced inconsistent result after apply │ │ When applying changes to googleplay_app_iam.quest_xboxbedrock, provider │ "provider[\"registry.terraform.io/oliver-binns/googleplay\"]" produced an │ unexpected new value: .app_id: was cty.StringVal("4972435805279727996"), │ but now cty.StringVal("quest.cmu.twa"). │ │ This is a bug in the provider, which should be reported in the provider's │ own issue tracker. ╵ ╷ │ Error: Provider produced inconsistent result after apply │ │ When applying changes to googleplay_app_iam.terrier_dkou, provider │ "provider[\"registry.terraform.io/oliver-binns/googleplay\"]" produced an │ unexpected new value: .app_id: was cty.StringVal("4975152612825535266"), │ but now cty.StringVal("build.terrier.app"). │ │ This is a bug in the provider, which should be reported in the provider's │ own issue tracker. ╵ ╷ │ Error: Provider produced inconsistent result after apply │ │ When applying changes to googleplay_app_iam.quest_kechezon, provider │ "provider[\"registry.terraform.io/oliver-binns/googleplay\"]" produced an │ unexpected new value: .app_id: was cty.StringVal("4972435805279727996"), │ but now cty.StringVal("quest.cmu.twa"). │ │ This is a bug in the provider, which should be reported in the provider's │ own issue tracker. ╵ ``` </details> --- ### 6. project: `sentry` dir: `tofu/sentry` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.37s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ sentry_team.cmu-eats: Creating... sentry_team.cmu-eats: Creation complete after 1s [id=cmu-eats] sentry_project.cmu-eats_eats: Creating... sentry_project.cmu-eats_eats: Creation complete after 1s [id=eats] sentry_key.cmu-eats_eats: Creating... sentry_key.cmu-eats_eats: Creation complete after 1s [id=46a2c57055681e48cc1891f4168268af] vault_kv_secret_v2.cmu-eats_eats_sentry_dsn: Creating... vault_kv_secret_v2.cmu-eats_eats_sentry_dsn: Creation complete after 0s [id=secret/data/secretspec/eats/prod/SENTRY_DSN] Apply complete! Resources: 4 added, 0 changed, 0 destroyed. ``` </details> --- ### 7. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.36s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ vault_kv_secret_v2.eats_staging_oauth_relay_url: Creating... vault_kv_secret_v2.eats_prod_oauth_relay_url: Creating... vault_kv_secret_v2.eats_preview_keycloak_url: Creating... vault_kv_secret_v2.eats_dev_keycloak_url: Creating... vault_kv_secret_v2.eats_prod_project_group: Creating... vault_kv_secret_v2.eats_dev_oauth_relay_url: Creating... vault_kv_secret_v2.eats_staging_keycloak_realm: Creating... vault_kv_secret_v2.eats_staging_project_admin_group: Creating... vault_kv_secret_v2.eats_staging_project_group: Creating... vault_kv_secret_v2.eats_prod_keycloak_url: Creating... vault_kv_secret_v2.eats_dev_keycloak_url: Creation complete after 0s [id=secret/data/secretspec/eats/dev/KEYCLOAK_URL] vault_kv_secret_v2.eats_staging_oauth_relay_url: Creation complete after 0s [id=secret/data/secretspec/eats/staging/OAUTH_RELAY_URL] vault_kv_secret_v2.eats_dev_oauth_relay_url: Creation complete after 0s [id=secret/data/secretspec/eats/dev/OAUTH_RELAY_URL] vault_kv_secret_v2.eats_prod_oauth_relay_url: Creation complete after 0s [id=secret/data/secretspec/eats/prod/OAUTH_RELAY_URL] vault_kv_secret_v2.eats_staging_project_admin_group: Creation complete after 0s [id=secret/data/secretspec/eats/staging/PROJECT_ADMIN_GROUP] vault_kv_secret_v2.eats_preview_oauth_relay_url: Creating... vault_kv_secret_v2.eats_prod_keycloak_realm: Creating... vault_kv_secret_v2.eats_preview_project_group: Creating... vault_kv_secret_v2.eats_staging_keycloak_url: Creating... vault_kv_secret_v2.eats_dev_project_admin_group: Creating... vault_kv_secret_v2.eats_staging_keycloak_realm: Creation complete after 0s [id=secret/data/secretspec/eats/staging/KEYCLOAK_REALM] vault_kv_secret_v2.eats_preview_keycloak_url: Creation complete after 0s [id=secret/data/secretspec/eats/preview/KEYCLOAK_URL] vault_kv_secret_v2.eats_dev_project_group: Creating... vault_kv_secret_v2.eats_prod_project_group: Creation complete after 0s [id=secret/data/secretspec/eats/prod/PROJECT_GROUP] vault_kv_secret_v2.eats_dev_keycloak_realm: Creating... vault_kv_secret_v2.eats_staging_project_group: Creation complete after 0s [id=secret/data/secretspec/eats/staging/PROJECT_GROUP] vault_kv_secret_v2.eats_prod_keycloak_url: Creation complete after 0s [id=secret/data/secretspec/eats/prod/KEYCLOAK_URL] vault_kv_secret_v2.eats_preview_project_admin_group: Creating... vault_kv_secret_v2.eats_prod_project_admin_group: Creating... vault_kv_secret_v2.eats_preview_keycloak_realm: Creating... vault_kv_secret_v2.eats_prod_keycloak_realm: Creation complete after 1s [id=secret/data/secretspec/eats/prod/KEYCLOAK_REALM] vault_kv_secret_v2.eats_staging_keycloak_url: Creation complete after 1s [id=secret/data/secretspec/eats/staging/KEYCLOAK_URL] vault_kv_secret_v2.eats_preview_oauth_relay_url: Creation complete after 1s [id=secret/data/secretspec/eats/preview/OAUTH_RELAY_URL] vault_kv_secret_v2.eats_preview_project_admin_group: Creation complete after 1s [id=secret/data/secretspec/eats/preview/PROJECT_ADMIN_GROUP] vault_kv_secret_v2.eats_preview_project_group: Creation complete after 1s [id=secret/data/secretspec/eats/preview/PROJECT_GROUP] vault_kv_secret_v2.eats_dev_keycloak_realm: Creation complete after 1s [id=secret/data/secretspec/eats/dev/KEYCLOAK_REALM] keycloak_openid_client.eats: Creating... vault_kv_secret_v2.eats_preview_keycloak_realm: Creation complete after 1s [id=secret/data/secretspec/eats/preview/KEYCLOAK_REALM] vault_kv_secret_v2.eats_prod_project_admin_group: Creation complete after 1s [id=secret/data/secretspec/eats/prod/PROJECT_ADMIN_GROUP] keycloak_openid_client.eats_staging: Creating... keycloak_group.project_cmu-eats_eats: Creating... keycloak_openid_client.eats_dev: Creating... vault_kv_secret_v2.eats_dev_project_group: Creation complete after 1s [id=secret/data/secretspec/eats/dev/PROJECT_GROUP] vault_kv_secret_v2.eats_dev_project_admin_group: Creation complete after 1s [id=secret/data/secretspec/eats/dev/PROJECT_ADMIN_GROUP] keycloak_group.project_cmu-eats_eats: Creation complete after 0s [id=41a1f091-ff21-421d-b98a-085de9cd77ba] keycloak_group.project_cmu-eats_eats_admins: Creating... keycloak_group_memberships.project_cmu-eats_eats: Creating... keycloak_openid_client.eats_staging: Creation complete after 0s [id=a17ef8d4-bd16-47d0-b10d-0efb1e800236] keycloak_openid_group_membership_protocol_mapper.eats_staging: Creating... vault_kv_secret_v2.eats_staging_oidc_client_id: Creating... vault_kv_secret_v2.eats_staging_oidc_client_secret: Creating... vault_kv_secret_v2.eats_preview_oidc_client_id: Creating... vault_kv_secret_v2.eats_preview_oidc_client_secret: Creating... keycloak_openid_client.eats_dev: Creation complete after 0s [id=696328b5-e9b0-4331-8442-5ffbe79aba7f] keycloak_openid_group_membership_protocol_mapper.eats_dev: Creating... vault_kv_secret_v2.eats_dev_oidc_client_secret: Creating... keycloak_openid_client.eats: Creation complete after 0s [id=fe3cf80b-e521-42a1-aff4-164cca377dd1] vault_kv_secret_v2.eats_dev_oidc_client_id: Creating... keycloak_group.project_cmu-eats_eats_admins: Creation complete after 0s [id=1a583ab1-469d-4eb4-a5a8-f2fd143ba829] vault_kv_secret_v2.eats_staging_oidc_client_secret: Creation complete after 0s [id=secret/data/secretspec/eats/staging/OIDC_CLIENT_SECRET] vault_kv_secret_v2.eats_prod_oidc_client_id: Creating... keycloak_openid_group_membership_protocol_mapper.eats: Creating... vault_kv_secret_v2.eats_staging_oidc_client_id: Creation complete after 0s [id=secret/data/secretspec/eats/staging/OIDC_CLIENT_ID] vault_kv_secret_v2.eats_prod_oidc_client_secret: Creating... vault_kv_secret_v2.eats_preview_oidc_client_secret: Creation complete after 0s [id=secret/data/secretspec/eats/preview/OIDC_CLIENT_SECRET] vault_kv_secret_v2.eats_preview_oidc_client_id: Creation complete after 0s [id=secret/data/secretspec/eats/preview/OIDC_CLIENT_ID] keycloak_group_memberships.project_cmu-eats_eats_admins: Creating... keycloak_openid_group_membership_protocol_mapper.eats_staging: Creation complete after 0s [id=1c2c928c-1af7-40ea-a10e-b654a0f35e4e] vault_kv_secret_v2.eats_dev_oidc_client_secret: Creation complete after 0s [id=secret/data/secretspec/eats/dev/OIDC_CLIENT_SECRET] keycloak_openid_group_membership_protocol_mapper.eats_dev: Creation complete after 0s [id=c537d1d2-f898-4a7a-ac30-d313657cdcb2] vault_kv_secret_v2.eats_dev_oidc_client_id: Creation complete after 0s [id=secret/data/secretspec/eats/dev/OIDC_CLIENT_ID] vault_kv_secret_v2.eats_prod_oidc_client_id: Creation complete after 0s [id=secret/data/secretspec/eats/prod/OIDC_CLIENT_ID] vault_kv_secret_v2.eats_prod_oidc_client_secret: Creation complete after 0s [id=secret/data/secretspec/eats/prod/OIDC_CLIENT_SECRET] keycloak_openid_group_membership_protocol_mapper.eats: Creation complete after 0s [id=d461ed68-2993-459b-9f87-f22d761eb29e] keycloak_group_memberships.project_cmu-eats_eats: Creation complete after 3s [id=scottylabs/group-memberships/41a1f091-ff21-421d-b98a-085de9cd77ba] keycloak_group_memberships.project_cmu-eats_eats_admins: Creation complete after 3s [id=scottylabs/group-memberships/1a583ab1-469d-4eb4-a5a8-f2fd143ba829] Apply complete! Resources: 38 added, 0 changed, 0 destroyed. ``` </details> --- ### 8. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.32s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ null_resource.slack_bridge_login_cmu_eats: Creating... null_resource.slack_bridge_login_cmu_eats: Provisioning with 'local-exec'... null_resource.slack_bridge_login_cmu_eats (local-exec): Executing: ["/bin/sh" "-c" "governance slack-join --channel C043UV30DHA"] null_resource.slack_bridge_login_cmu_eats: Creation complete after 0s [id=698518873962630849] synapse_mautrix_slack_link.cmu_eats: Creating... synapse_mautrix_slack_link.cmu_eats: Still creating... [10s elapsed] synapse_mautrix_slack_link.cmu_eats: Creation complete after 17s [id=1541336561105240134] Apply complete! Resources: 2 added, 0 changed, 0 destroyed. ``` </details> --- ### 9. project: `git` dir: `tofu/git` workspace: `default` **Apply Error** <details><summary>Show Output</summary> ``` running 'sh -c' '/nix/store/9dj5sdaq6vksi96ibn9vpyzjcbhf26jv-opentofu-1.12.5/bin/tofu apply -input=false "/var/lib/atlantis/repos/ScottyLabs/governance/178/default/tofu/git/git-default.tfplan"' in '/var/lib/atlantis/repos/ScottyLabs/governance/178/default/tofu/git': exit status 1 Finished `release` profile [optimized] target(s) in 0.13s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ github_team_membership.quest_brightz: Creating... github_team.cmu-eats: Creating... github_repository.cmu-eats_eats: Creating... github_repository.slai_cmugpt_agent: Modifying... [id=cmugpt-agent] github_repository.slai_cmugpt_surface: Modifying... [id=cmugpt-surface] github_repository_ruleset.devops_kennel_ruleset: Modifying... [id=13193453] github_repository_ruleset.cbp_bus_sign_ruleset: Modifying... [id=15236550] github_repository_ruleset.cbp_discord_verify_ruleset: Modifying... [id=11449154] github_repository_ruleset.devops_infrastructure_ruleset: Modifying... [id=11275118] github_repository_ruleset.uia_scottylabs_org_ruleset: Modifying... [id=18112214] forgejo_repository.cmu-eats_eats: Creating... forgejo_repository.cmu-eats_eats: Creation complete after 1s [name=eats] forgejo_team.eats_leads: Creating... forgejo_team.eats_leads: Creation complete after 0s [name=eats-leads] forgejo_team.cmu-eats: Creating... forgejo_team.cmu-eats: Creation complete after 0s [name=cmu-eats] restapi_object.cmu-eats_eats_push_mirror: Creating... restapi_object.cmu-eats_eats_push_mirror: Creation complete after 0s [id=remote_mirror_KOBYPLZTKSZ] restapi_object.cmu-eats_eats_kennel_webhook: Creating... restapi_object.cmu-eats_eats_kennel_webhook: Creation complete after 0s [id=30] github_repository_ruleset.quest_quest_ruleset: Modifying... [id=15750052] github_team_membership.quest_brightz: Creation complete after 9s [id=17330197:bright-zheng] github_repository_ruleset.devops_governance_ruleset: Modifying... [id=14488792] github_repository.cmu-eats_eats: Still creating... [10s elapsed] github_repository.slai_cmugpt_agent: Still modifying... [id=cmugpt-agent, 10s elapsed] github_repository.slai_cmugpt_surface: Still modifying... [id=cmugpt-surface, 10s elapsed] github_repository_ruleset.devops_kennel_ruleset: Still modifying... [id=13193453, 10s elapsed] github_repository_ruleset.cbp_bus_sign_ruleset: Still modifying... [id=15236550, 10s elapsed] github_repository_ruleset.cbp_discord_verify_ruleset: Still modifying... [id=11449154, 10s elapsed] github_repository_ruleset.devops_infrastructure_ruleset: Still modifying... [id=11275118, 10s elapsed] github_repository_ruleset.uia_scottylabs_org_ruleset: Still modifying... [id=18112214, 10s elapsed] github_repository.slai_cmugpt_agent: Modifications complete after 11s [id=cmugpt-agent] github_repository_ruleset.devops_collie_ruleset: Modifying... [id=18865204] github_repository.slai_cmugpt_surface: Modifications complete after 12s [id=cmugpt-surface] github_repository_ruleset.cmu-maps_maps_ruleset: Modifying... [id=18864961] github_repository_ruleset.quest_quest_ruleset: Still modifying... [id=15750052, 10s elapsed] github_repository_ruleset.devops_kennel_ruleset: Modifications complete after 12s [id=13193453] github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Modifying... [id=16848966] github_repository_ruleset.cbp_discord_verify_ruleset: Modifications complete after 14s [id=11449154] github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Modifying... [id=17653622] github_repository_ruleset.cbp_bus_sign_ruleset: Modifications complete after 15s [id=15236550] github_repository_ruleset.cbp_dalmatian_ruleset: Modifying... [id=17238915] github_repository_ruleset.devops_infrastructure_ruleset: Modifications complete after 16s [id=11275118] github_repository_ruleset.devops_documentation_ruleset: Modifying... [id=17079368] github_repository_ruleset.uia_scottylabs_org_ruleset: Modifications complete after 18s [id=18112214] github_repository_ruleset.cmu-housing_housing_ruleset: Modifying... [id=17421604] github_repository_ruleset.devops_governance_ruleset: Still modifying... [id=14488792, 10s elapsed] github_repository_ruleset.quest_quest_ruleset: Modifications complete after 17s [id=15750052] github_repository_ruleset.cmu-courses_internet_archive_ruleset: Modifying... [id=11988074] github_repository.cmu-eats_eats: Still creating... [20s elapsed] github_repository_ruleset.devops_collie_ruleset: Still modifying... [id=18865204, 10s elapsed] github_repository_ruleset.cmu-maps_maps_ruleset: Still modifying... [id=18864961, 10s elapsed] github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Still modifying... [id=16848966, 10s elapsed] github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Still modifying... [id=17653622, 10s elapsed] github_repository_ruleset.devops_governance_ruleset: Modifications complete after 15s [id=14488792] github_repository_ruleset.cmu-courses_courses_ruleset: Modifying... [id=15855161] github_repository_ruleset.cbp_dalmatian_ruleset: Still modifying... [id=17238915, 10s elapsed] github_repository_ruleset.devops_collie_ruleset: Modifications complete after 15s [id=18865204] forgejo_team_repository.eats_leads_eats: Creating... forgejo_team_repository.eats_leads_eats: Creation complete after 0s forgejo_team_member.eats_leads_anish: Creating... forgejo_team_member.eats_leads_anish: Creation complete after 0s github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Modifying... [id=18920618] github_repository_ruleset.devops_documentation_ruleset: Still modifying... [id=17079368, 10s elapsed] github_repository_ruleset.cmu-maps_maps_ruleset: Modifications complete after 15s [id=18864961] github_repository_ruleset.slai_mcp_server_ruleset: Modifying... [id=17082154] github_repository_ruleset.cmu-housing_housing_ruleset: Still modifying... [id=17421604, 10s elapsed] github_repository_ruleset.tartan-vote_tartan_vote_ruleset: Modifications complete after 17s [id=16848966] github_repository_ruleset.terrier_terrier_ruleset: Modifying... [id=12671003] github_repository_ruleset.cmu-courses_internet_archive_ruleset: Still modifying... [id=11988074, 10s elapsed] github_repository.cmu-eats_eats: Still creating... [30s elapsed] github_repository_ruleset.slai_cmugpt_sms_surface_ruleset: Modifications complete after 16s [id=17653622] forgejo_team_member.cmu-eats_anish: Creating... forgejo_team_member.cmu-eats_anish: Creation complete after 0s forgejo_team_repository.cmu-eats_eats: Creating... forgejo_team_repository.cmu-eats_eats: Creation complete after 0s github_repository_ruleset.cbp_groupme_mirror_ruleset: Modifying... [id=12503296] github_repository_ruleset.cbp_dalmatian_ruleset: Modifications complete after 17s [id=17238915] github_repository_ruleset.devops_link_shortener_ruleset: Modifying... [id=17911647] github_repository_ruleset.devops_documentation_ruleset: Modifications complete after 17s [id=17079368] github_repository_ruleset.uia_components_ruleset: Modifying... [id=16366085] github_repository_ruleset.cmu-courses_courses_ruleset: Still modifying... [id=15855161, 10s elapsed] github_repository_ruleset.cmu-housing_housing_ruleset: Modifications complete after 17s [id=17421604] github_repository_ruleset.cmu-cal_cal_ruleset: Modifying... [id=18919281] github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Still modifying... [id=18920618, 10s elapsed] github_repository_ruleset.cmu-courses_internet_archive_ruleset: Modifications complete after 17s [id=11988074] github_repository_ruleset.slai_cmugpt_agent_ruleset: Modifying... [id=16476661] github_repository_ruleset.slai_mcp_server_ruleset: Still modifying... [id=17082154, 10s elapsed] github_repository.cmu-eats_eats: Creation complete after 37s [id=eats] github_repository_ruleset.slai_cmugpt_surface_ruleset: Modifying... [id=16476662] github_repository_ruleset.terrier_terrier_ruleset: Still modifying... [id=12671003, 10s elapsed] github_repository_ruleset.cmu-courses_courses_ruleset: Modifications complete after 16s [id=15855161] github_actions_repository_permissions.cmu-eats_eats_actions: Creating... github_repository_ruleset.cbp_groupme_mirror_ruleset: Still modifying... [id=12503296, 10s elapsed] github_repository_ruleset.devops_link_shortener_ruleset: Still modifying... [id=17911647, 10s elapsed] github_repository_ruleset.lost-and-found_lost_and_found_v2_ruleset: Modifications complete after 16s [id=18920618] github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Creating... github_repository_ruleset.slai_mcp_server_ruleset: Modifications complete after 16s [id=17082154] github_repository_ruleset.uia_components_ruleset: Still modifying... [id=16366085, 10s elapsed] github_repository_ruleset.terrier_terrier_ruleset: Modifications complete after 15s [id=12671003] github_repository_ruleset.cmu-cal_cal_ruleset: Still modifying... [id=18919281, 10s elapsed] github_repository_ruleset.cbp_groupme_mirror_ruleset: Modifications complete after 16s [id=12503296] github_repository_ruleset.slai_cmugpt_agent_ruleset: Still modifying... [id=16476661, 10s elapsed] github_repository_ruleset.slai_cmugpt_surface_ruleset: Still modifying... [id=16476662, 10s elapsed] github_repository_ruleset.devops_link_shortener_ruleset: Modifications complete after 15s [id=17911647] github_repository_ruleset.uia_components_ruleset: Modifications complete after 16s [id=16366085] github_actions_repository_permissions.cmu-eats_eats_actions: Still creating... [10s elapsed] github_repository_ruleset.cmu-cal_cal_ruleset: Modifications complete after 15s [id=18919281] github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Still creating... [10s elapsed] github_repository_ruleset.slai_cmugpt_agent_ruleset: Modifications complete after 17s [id=16476661] github_repository_ruleset.slai_cmugpt_surface_ruleset: Modifications complete after 17s [id=16476662] github_actions_repository_permissions.cmu-eats_eats_actions: Creation complete after 17s [id=eats] github_repository_deploy_key.cmu-eats_eats_mirror_deploy_key: Creation complete after 15s [id=eats:161201842] github_repository_ruleset.cmu-eats_eats_ruleset: Creating... github_repository_ruleset.cmu-eats_eats_ruleset: Creation complete after 0s [id=21330218] ╷ │ Error: POST https://api.github.com/orgs/ScottyLabs/teams: 422 Validation Failed [{Resource:Team Field:data Code:unprocessable Message:Name must be unique for this org}] │ │ with github_team.cmu-eats, │ on github_teams.tf.json line 19, in resource.github_team.cmu-eats: │ 19: }, │ ╵ ``` </details> --- ### Apply Summary 9 projects, 7 successful, 0 failed, 2 errored
fix: change app ids to actual strings
Some checks failed
CI / build (pull_request) Has been skipped
kennel/build skipped: no services or static sites defined
atlantis/plan: slack No changes. Your infrastructure matches the configuration.
atlantis/plan: posthog No changes. Your infrastructure matches the configuration.
atlantis/plan: openbao No changes. Your infrastructure matches the configuration.
atlantis/plan: matrix_bridges No changes. Your infrastructure matches the configuration.
atlantis/plan: sentry No changes. Your infrastructure matches the configuration.
atlantis/plan: discord No changes. Your infrastructure matches the configuration.
atlantis/plan: keycloak No changes. Your infrastructure matches the configuration.
atlantis/plan: git Plan: 1 to add, 0 to change, 0 to destroy.
atlantis/apply: discord Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: posthog Apply succeeded.
atlantis/apply: sentry Apply succeeded.
atlantis/apply: openbao Apply succeeded.
atlantis/apply: matrix_bridges Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply: git Apply succeeded.
Validate / Governance (pull_request_target) Successful in 4m49s
CI / check-1 (pull_request) Successful in 5m59s
CI / check (pull_request) Successful in 0s
atlantis/plan: google_play Plan: 5 to add, 0 to change, 0 to destroy.
atlantis/plan 9/9 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: google_play Apply failed.
atlantis/apply 1/9 projects failed to apply.
bca92d47cd
Signed-off-by: Anish Pallati <i@anish.land>

Ran Plan for 9 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: openbao dir: tofu/openbao workspace: default
  3. project: google_play dir: tofu/google_play workspace: default
  4. project: matrix_bridges dir: tofu/matrix_bridges workspace: default
  5. project: posthog dir: tofu/posthog workspace: default
  6. project: sentry dir: tofu/sentry workspace: default
  7. project: discord dir: tofu/discord workspace: default
  8. project: keycloak dir: tofu/keycloak workspace: default
  9. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.25s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

2. project: openbao dir: tofu/openbao workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.24s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p openbao
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p openbao
    

3. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.48s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
-/+ destroy and then create replacement

OpenTofu will perform the following actions:

  # googleplay_app_iam.quest_brightz is tainted, so it must be replaced
-/+ resource "googleplay_app_iam" "quest_brightz" {
!       expanded_permissions = [
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> (known after apply)
        # (3 unchanged attributes hidden)
    }

  # googleplay_app_iam.quest_kechezon is tainted, so it must be replaced
-/+ resource "googleplay_app_iam" "quest_kechezon" {
!       expanded_permissions = [
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> (known after apply)
        # (3 unchanged attributes hidden)
    }

  # googleplay_app_iam.quest_xboxbedrock is tainted, so it must be replaced
-/+ resource "googleplay_app_iam" "quest_xboxbedrock" {
!       expanded_permissions = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> (known after apply)
        # (3 unchanged attributes hidden)
    }

  # googleplay_app_iam.terrier_dkou is tainted, so it must be replaced
-/+ resource "googleplay_app_iam" "terrier_dkou" {
!       expanded_permissions = [
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> (known after apply)
        # (3 unchanged attributes hidden)
    }

  # googleplay_app_iam.terrier_kothurvidur is tainted, so it must be replaced
-/+ resource "googleplay_app_iam" "terrier_kothurvidur" {
!       expanded_permissions = [
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> (known after apply)
        # (3 unchanged attributes hidden)
    }

Plan: 5 to add, 0 to change, 5 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 5 to add, 0 to change, 5 to destroy.


4. project: matrix_bridges dir: tofu/matrix_bridges workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.44s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p matrix_bridges
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p matrix_bridges
    

5. project: posthog dir: tofu/posthog workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.45s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p posthog
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p posthog
    

6. project: sentry dir: tofu/sentry workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.35s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p sentry
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p sentry
    

7. project: discord dir: tofu/discord workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on shared package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.45s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

8. project: keycloak dir: tofu/keycloak workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.30s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

9. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.20s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # github_team.cmu-eats will be created
+ resource "github_team" "cmu-eats" {
+       create_default_maintainer = false
+       etag                      = (known after apply)
+       id                        = (known after apply)
+       members_count             = (known after apply)
+       name                      = "CMU Eats"
+       node_id                   = (known after apply)
+       notification_setting      = "notifications_enabled"
+       parent_team_read_id       = (known after apply)
+       parent_team_read_slug     = (known after apply)
+       privacy                   = "closed"
+       slug                      = (known after apply)
    }

  # github_team_membership.cmu-eats_anish will be created
+ resource "github_team_membership" "cmu-eats_anish" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "maintainer"
+       team_id  = (known after apply)
+       username = "ap-1"
    }

Plan: 2 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 2 to add, 0 to change, 0 to destroy.


Plan Summary

9 projects, 2 with changes, 7 with no changes, 0 failed

  • To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 9 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `openbao` dir: `tofu/openbao` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` 1. project: `posthog` dir: `tofu/posthog` workspace: `default` 1. project: `sentry` dir: `tofu/sentry` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.25s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` --- ### 2. project: `openbao` dir: `tofu/openbao` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.24s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p openbao ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fopenbao%252Fdefault%252Fopenbao) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p openbao ``` --- ### 3. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.48s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: -/+ destroy and then create replacement OpenTofu will perform the following actions: # googleplay_app_iam.quest_brightz is tainted, so it must be replaced -/+ resource "googleplay_app_iam" "quest_brightz" { ! expanded_permissions = [ - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> (known after apply) # (3 unchanged attributes hidden) } # googleplay_app_iam.quest_kechezon is tainted, so it must be replaced -/+ resource "googleplay_app_iam" "quest_kechezon" { ! expanded_permissions = [ - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> (known after apply) # (3 unchanged attributes hidden) } # googleplay_app_iam.quest_xboxbedrock is tainted, so it must be replaced -/+ resource "googleplay_app_iam" "quest_xboxbedrock" { ! expanded_permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> (known after apply) # (3 unchanged attributes hidden) } # googleplay_app_iam.terrier_dkou is tainted, so it must be replaced -/+ resource "googleplay_app_iam" "terrier_dkou" { ! expanded_permissions = [ - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> (known after apply) # (3 unchanged attributes hidden) } # googleplay_app_iam.terrier_kothurvidur is tainted, so it must be replaced -/+ resource "googleplay_app_iam" "terrier_kothurvidur" { ! expanded_permissions = [ - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> (known after apply) # (3 unchanged attributes hidden) } Plan: 5 to add, 0 to change, 5 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 5 to add, 0 to change, 5 to destroy. --- ### 4. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.44s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p matrix_bridges ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fmatrix_bridges%252Fdefault%252Fmatrix_bridges) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p matrix_bridges ``` --- ### 5. project: `posthog` dir: `tofu/posthog` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.45s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p posthog ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fposthog%252Fdefault%252Fposthog) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p posthog ``` --- ### 6. project: `sentry` dir: `tofu/sentry` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.35s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p sentry ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fsentry%252Fdefault%252Fsentry) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p sentry ``` --- ### 7. project: `discord` dir: `tofu/discord` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on shared package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.45s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` --- ### 8. project: `keycloak` dir: `tofu/keycloak` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.30s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` --- ### 9. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.20s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # github_team.cmu-eats will be created + resource "github_team" "cmu-eats" { + create_default_maintainer = false + etag = (known after apply) + id = (known after apply) + members_count = (known after apply) + name = "CMU Eats" + node_id = (known after apply) + notification_setting = "notifications_enabled" + parent_team_read_id = (known after apply) + parent_team_read_slug = (known after apply) + privacy = "closed" + slug = (known after apply) } # github_team_membership.cmu-eats_anish will be created + resource "github_team_membership" "cmu-eats_anish" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = (known after apply) + username = "ap-1" } Plan: 2 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 2 to add, 0 to change, 0 to destroy. --- ### Plan Summary 9 projects, 2 with changes, 7 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Author
Owner

atlantis import -p git github_team.cmu-eats 16807600

atlantis import -p git github_team.cmu-eats 16807600

Ran Import for project: git dir: tofu/git workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.05s

data.external.identity_dkou: Reading...
data.external.identity_maybe_yiyi: Reading...
data.external.identity_liamwalk: Reading...
data.external.identity_riztofu: Reading...
data.external.identity_anish: Reading...
data.external.identity_krishsax: Reading...
data.external.identity_kechezon: Reading...
data.external.identity_tentype: Reading...
data.external.identity_rzhu3: Reading...
data.external.identity_krishsax: Read complete after 0s [id=-]
data.external.identity_thesuperrl: Reading...
data.external.identity_dkou: Read complete after 0s [id=-]
data.external.identity_xboxbedrock: Reading...
data.external.identity_liamwalk: Read complete after 0s [id=-]
data.external.identity_jiyas: Reading...
data.external.identity_maybe_yiyi: Read complete after 0s [id=-]
data.external.identity_jc_965: Reading...
data.external.identity_kechezon: Read complete after 0s [id=-]
data.external.identity_kothurvidur: Reading...
data.external.identity_anish: Read complete after 0s [id=-]
data.external.identity_tentype: Read complete after 0s [id=-]
data.external.identity_brightz: Reading...
data.external.identity_rzhu3: Read complete after 0s [id=-]
data.external.identity_riztofu: Read complete after 0s [id=-]
data.forgejo_organization.this: Reading...
data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs]
github_team.cmu-eats: Importing from ID "16807600"...
github_team.cmu-eats: Import prepared!
  Prepared github_team for import
data.external.identity_xboxbedrock: Read complete after 0s [id=-]
github_team.cmu-eats: Refreshing state... [id=16807600]
data.external.identity_thesuperrl: Read complete after 0s [id=-]
data.external.identity_kothurvidur: Read complete after 0s [id=-]
data.external.identity_jc_965: Read complete after 0s [id=-]
data.external.identity_brightz: Read complete after 0s [id=-]
data.external.identity_jiyas: Read complete after 0s [id=-]

Import successful!

The resources that were imported are shown above. These resources are now in
your OpenTofu state and will henceforth be managed by OpenTofu.
🚮 A plan file was discarded. Re-plan would be required before applying.
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    
Ran Import for project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.05s data.external.identity_dkou: Reading... data.external.identity_maybe_yiyi: Reading... data.external.identity_liamwalk: Reading... data.external.identity_riztofu: Reading... data.external.identity_anish: Reading... data.external.identity_krishsax: Reading... data.external.identity_kechezon: Reading... data.external.identity_tentype: Reading... data.external.identity_rzhu3: Reading... data.external.identity_krishsax: Read complete after 0s [id=-] data.external.identity_thesuperrl: Reading... data.external.identity_dkou: Read complete after 0s [id=-] data.external.identity_xboxbedrock: Reading... data.external.identity_liamwalk: Read complete after 0s [id=-] data.external.identity_jiyas: Reading... data.external.identity_maybe_yiyi: Read complete after 0s [id=-] data.external.identity_jc_965: Reading... data.external.identity_kechezon: Read complete after 0s [id=-] data.external.identity_kothurvidur: Reading... data.external.identity_anish: Read complete after 0s [id=-] data.external.identity_tentype: Read complete after 0s [id=-] data.external.identity_brightz: Reading... data.external.identity_rzhu3: Read complete after 0s [id=-] data.external.identity_riztofu: Read complete after 0s [id=-] data.forgejo_organization.this: Reading... data.forgejo_organization.this: Read complete after 0s [name=ScottyLabs] github_team.cmu-eats: Importing from ID "16807600"... github_team.cmu-eats: Import prepared! Prepared github_team for import data.external.identity_xboxbedrock: Read complete after 0s [id=-] github_team.cmu-eats: Refreshing state... [id=16807600] data.external.identity_thesuperrl: Read complete after 0s [id=-] data.external.identity_kothurvidur: Read complete after 0s [id=-] data.external.identity_jc_965: Read complete after 0s [id=-] data.external.identity_brightz: Read complete after 0s [id=-] data.external.identity_jiyas: Read complete after 0s [id=-] Import successful! The resources that were imported are shown above. These resources are now in your OpenTofu state and will henceforth be managed by OpenTofu. ``` </details> :put_litter_in_its_place: A plan file was discarded. Re-plan would be required before applying. * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ```
Author
Owner

atlantis plan

atlantis plan

Ran Plan for 9 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: posthog dir: tofu/posthog workspace: default
  3. project: google_play dir: tofu/google_play workspace: default
  4. project: openbao dir: tofu/openbao workspace: default
  5. project: matrix_bridges dir: tofu/matrix_bridges workspace: default
  6. project: sentry dir: tofu/sentry workspace: default
  7. project: discord dir: tofu/discord workspace: default
  8. project: keycloak dir: tofu/keycloak workspace: default
  9. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.28s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

2. project: posthog dir: tofu/posthog workspace: default

Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.21s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p posthog
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p posthog
    

3. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.33s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
-/+ destroy and then create replacement

OpenTofu will perform the following actions:

  # googleplay_app_iam.quest_brightz is tainted, so it must be replaced
-/+ resource "googleplay_app_iam" "quest_brightz" {
!       expanded_permissions = [
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> (known after apply)
        # (3 unchanged attributes hidden)
    }

  # googleplay_app_iam.quest_kechezon is tainted, so it must be replaced
-/+ resource "googleplay_app_iam" "quest_kechezon" {
!       expanded_permissions = [
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> (known after apply)
        # (3 unchanged attributes hidden)
    }

  # googleplay_app_iam.quest_xboxbedrock is tainted, so it must be replaced
-/+ resource "googleplay_app_iam" "quest_xboxbedrock" {
!       expanded_permissions = [
-           "CAN_MANAGE_APP_CONTENT",
-           "CAN_MANAGE_PUBLIC_APKS",
-           "CAN_MANAGE_PUBLIC_LISTING",
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_REPLY_TO_REVIEWS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> (known after apply)
        # (3 unchanged attributes hidden)
    }

  # googleplay_app_iam.terrier_dkou is tainted, so it must be replaced
-/+ resource "googleplay_app_iam" "terrier_dkou" {
!       expanded_permissions = [
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> (known after apply)
        # (3 unchanged attributes hidden)
    }

  # googleplay_app_iam.terrier_kothurvidur is tainted, so it must be replaced
-/+ resource "googleplay_app_iam" "terrier_kothurvidur" {
!       expanded_permissions = [
-           "CAN_MANAGE_TRACK_APKS",
-           "CAN_MANAGE_TRACK_USERS",
-           "CAN_VIEW_APP_QUALITY",
-           "CAN_VIEW_NON_FINANCIAL_DATA",
        ] -> (known after apply)
        # (3 unchanged attributes hidden)
    }

Plan: 5 to add, 0 to change, 5 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 5 to add, 0 to change, 5 to destroy.


4. project: openbao dir: tofu/openbao workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.36s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p openbao
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p openbao
    

5. project: matrix_bridges dir: tofu/matrix_bridges workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.39s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p matrix_bridges
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p matrix_bridges
    

6. project: sentry dir: tofu/sentry workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.32s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p sentry
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p sentry
    

7. project: discord dir: tofu/discord workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.21s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

8. project: keycloak dir: tofu/keycloak workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.34s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

9. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.22s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # github_team_membership.cmu-eats_anish will be created
+ resource "github_team_membership" "cmu-eats_anish" {
+       etag     = (known after apply)
+       id       = (known after apply)
+       role     = "maintainer"
+       team_id  = "16807600"
+       username = "ap-1"
    }

Plan: 1 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan: 1 to add, 0 to change, 0 to destroy.


Plan Summary

9 projects, 2 with changes, 7 with no changes, 0 failed

  • To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 9 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `posthog` dir: `tofu/posthog` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `openbao` dir: `tofu/openbao` workspace: `default` 1. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` 1. project: `sentry` dir: `tofu/sentry` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.28s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` --- ### 2. project: `posthog` dir: `tofu/posthog` workspace: `default` ```diff Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.21s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p posthog ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fposthog%252Fdefault%252Fposthog) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p posthog ``` --- ### 3. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.33s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: -/+ destroy and then create replacement OpenTofu will perform the following actions: # googleplay_app_iam.quest_brightz is tainted, so it must be replaced -/+ resource "googleplay_app_iam" "quest_brightz" { ! expanded_permissions = [ - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> (known after apply) # (3 unchanged attributes hidden) } # googleplay_app_iam.quest_kechezon is tainted, so it must be replaced -/+ resource "googleplay_app_iam" "quest_kechezon" { ! expanded_permissions = [ - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> (known after apply) # (3 unchanged attributes hidden) } # googleplay_app_iam.quest_xboxbedrock is tainted, so it must be replaced -/+ resource "googleplay_app_iam" "quest_xboxbedrock" { ! expanded_permissions = [ - "CAN_MANAGE_APP_CONTENT", - "CAN_MANAGE_PUBLIC_APKS", - "CAN_MANAGE_PUBLIC_LISTING", - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_REPLY_TO_REVIEWS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> (known after apply) # (3 unchanged attributes hidden) } # googleplay_app_iam.terrier_dkou is tainted, so it must be replaced -/+ resource "googleplay_app_iam" "terrier_dkou" { ! expanded_permissions = [ - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> (known after apply) # (3 unchanged attributes hidden) } # googleplay_app_iam.terrier_kothurvidur is tainted, so it must be replaced -/+ resource "googleplay_app_iam" "terrier_kothurvidur" { ! expanded_permissions = [ - "CAN_MANAGE_TRACK_APKS", - "CAN_MANAGE_TRACK_USERS", - "CAN_VIEW_APP_QUALITY", - "CAN_VIEW_NON_FINANCIAL_DATA", ] -> (known after apply) # (3 unchanged attributes hidden) } Plan: 5 to add, 0 to change, 5 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 5 to add, 0 to change, 5 to destroy. --- ### 4. project: `openbao` dir: `tofu/openbao` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.36s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p openbao ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fopenbao%252Fdefault%252Fopenbao) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p openbao ``` --- ### 5. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.39s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p matrix_bridges ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fmatrix_bridges%252Fdefault%252Fmatrix_bridges) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p matrix_bridges ``` --- ### 6. project: `sentry` dir: `tofu/sentry` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.32s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p sentry ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fsentry%252Fdefault%252Fsentry) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p sentry ``` --- ### 7. project: `discord` dir: `tofu/discord` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.21s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` --- ### 8. project: `keycloak` dir: `tofu/keycloak` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.34s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` --- ### 9. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.22s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # github_team_membership.cmu-eats_anish will be created + resource "github_team_membership" "cmu-eats_anish" { + etag = (known after apply) + id = (known after apply) + role = "maintainer" + team_id = "16807600" + username = "ap-1" } Plan: 1 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` Plan: 1 to add, 0 to change, 0 to destroy. --- ### Plan Summary 9 projects, 2 with changes, 7 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Author
Owner

atlantis apply

atlantis apply

Ran Apply for 9 projects:

  1. project: discord dir: tofu/discord workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: posthog dir: tofu/posthog workspace: default
  4. project: google_play dir: tofu/google_play workspace: default
  5. project: sentry dir: tofu/sentry workspace: default
  6. project: openbao dir: tofu/openbao workspace: default
  7. project: matrix_bridges dir: tofu/matrix_bridges workspace: default
  8. project: keycloak dir: tofu/keycloak workspace: default
  9. project: git dir: tofu/git workspace: default

1. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.34s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

2. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.43s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

3. project: posthog dir: tofu/posthog workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.26s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

4. project: google_play dir: tofu/google_play workspace: default

Apply Error

Show Output
running 'sh -c' '/nix/store/9dj5sdaq6vksi96ibn9vpyzjcbhf26jv-opentofu-1.12.5/bin/tofu apply -input=false "/var/lib/atlantis/repos/ScottyLabs/governance/178/default/tofu/google_play/google_play-default.tfplan"' in '/var/lib/atlantis/repos/ScottyLabs/governance/178/default/tofu/google_play': exit status 1
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.24s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
googleplay_app_iam.terrier_kothurvidur: Destroying...
googleplay_app_iam.quest_brightz: Destroying...
googleplay_app_iam.terrier_dkou: Destroying...
googleplay_app_iam.quest_kechezon: Destroying...
googleplay_app_iam.quest_xboxbedrock: Destroying...
╷
│ Error: Client Error
│ 
│ Unable to delete permission, got error: googleapi: Error 404: The requested
│ object was not found, notFound
╵
╷
│ Error: Client Error
│ 
│ Unable to delete permission, got error: googleapi: Error 404: The requested
│ object was not found, notFound
╵
╷
│ Error: Client Error
│ 
│ Unable to delete permission, got error: googleapi: Error 404: The requested
│ object was not found, notFound
╵
╷
│ Error: Client Error
│ 
│ Unable to delete permission, got error: googleapi: Error 404: The requested
│ object was not found, notFound
╵
╷
│ Error: Client Error
│ 
│ Unable to delete permission, got error: googleapi: Error 404: The requested
│ object was not found, notFound
╵


5. project: sentry dir: tofu/sentry workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.26s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

6. project: openbao dir: tofu/openbao workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.25s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

7. project: matrix_bridges dir: tofu/matrix_bridges workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.38s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

8. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.37s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

9. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.41s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
github_team_membership.cmu-eats_anish: Creating...
github_team_membership.cmu-eats_anish: Creation complete after 2s [id=16807600:ap-1]

Apply complete! Resources: 1 added, 0 changed, 0 destroyed.

Apply Summary

9 projects, 8 successful, 0 failed, 1 errored

Ran Apply for 9 projects: 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `posthog` dir: `tofu/posthog` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `sentry` dir: `tofu/sentry` workspace: `default` 1. project: `openbao` dir: `tofu/openbao` workspace: `default` 1. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.34s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.43s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `posthog` dir: `tofu/posthog` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.26s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `google_play` dir: `tofu/google_play` workspace: `default` **Apply Error** <details><summary>Show Output</summary> ``` running 'sh -c' '/nix/store/9dj5sdaq6vksi96ibn9vpyzjcbhf26jv-opentofu-1.12.5/bin/tofu apply -input=false "/var/lib/atlantis/repos/ScottyLabs/governance/178/default/tofu/google_play/google_play-default.tfplan"' in '/var/lib/atlantis/repos/ScottyLabs/governance/178/default/tofu/google_play': exit status 1 Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.24s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ googleplay_app_iam.terrier_kothurvidur: Destroying... googleplay_app_iam.quest_brightz: Destroying... googleplay_app_iam.terrier_dkou: Destroying... googleplay_app_iam.quest_kechezon: Destroying... googleplay_app_iam.quest_xboxbedrock: Destroying... ╷ │ Error: Client Error │ │ Unable to delete permission, got error: googleapi: Error 404: The requested │ object was not found, notFound ╵ ╷ │ Error: Client Error │ │ Unable to delete permission, got error: googleapi: Error 404: The requested │ object was not found, notFound ╵ ╷ │ Error: Client Error │ │ Unable to delete permission, got error: googleapi: Error 404: The requested │ object was not found, notFound ╵ ╷ │ Error: Client Error │ │ Unable to delete permission, got error: googleapi: Error 404: The requested │ object was not found, notFound ╵ ╷ │ Error: Client Error │ │ Unable to delete permission, got error: googleapi: Error 404: The requested │ object was not found, notFound ╵ ``` </details> --- ### 5. project: `sentry` dir: `tofu/sentry` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.26s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 6. project: `openbao` dir: `tofu/openbao` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.25s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 7. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.38s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 8. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.37s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 9. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.41s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ github_team_membership.cmu-eats_anish: Creating... github_team_membership.cmu-eats_anish: Creation complete after 2s [id=16807600:ap-1] Apply complete! Resources: 1 added, 0 changed, 0 destroyed. ``` </details> --- ### Apply Summary 9 projects, 8 successful, 0 failed, 1 errored
Author
Owner

atlantis plan -p google_play

atlantis plan -p google_play

Ran Plan for project: google_play dir: tofu/google_play workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.06s

data.external.identity_rzhu3: Read complete after 0s [id=-]

OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # googleplay_app_iam.quest_brightz will be created
+ resource "googleplay_app_iam" "quest_brightz" {
+       app_id               = "quest.cmu.twa"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "brightz@andrew.cmu.edu"
    }

  # googleplay_app_iam.quest_kechezon will be created
+ resource "googleplay_app_iam" "quest_kechezon" {
+       app_id               = "quest.cmu.twa"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "kechezon@andrew.cmu.edu"
    }

  # googleplay_app_iam.quest_xboxbedrock will be created
+ resource "googleplay_app_iam" "quest_xboxbedrock" {
+       app_id               = "quest.cmu.twa"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_APP_CONTENT",
+           "CAN_MANAGE_PUBLIC_APKS",
+           "CAN_MANAGE_PUBLIC_LISTING",
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_REPLY_TO_REVIEWS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "spola2@andrew.cmu.edu"
    }

  # googleplay_app_iam.terrier_dkou will be created
+ resource "googleplay_app_iam" "terrier_dkou" {
+       app_id               = "build.terrier.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "dkou@andrew.cmu.edu"
    }

  # googleplay_app_iam.terrier_kothurvidur will be created
+ resource "googleplay_app_iam" "terrier_kothurvidur" {
+       app_id               = "build.terrier.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "vkothur@andrew.cmu.edu"
    }

Plan: 5 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 5 to add, 0 to change, 0 to destroy.


  • To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.06s data.external.identity_rzhu3: Read complete after 0s [id=-] OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # googleplay_app_iam.quest_brightz will be created + resource "googleplay_app_iam" "quest_brightz" { + app_id = "quest.cmu.twa" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "brightz@andrew.cmu.edu" } # googleplay_app_iam.quest_kechezon will be created + resource "googleplay_app_iam" "quest_kechezon" { + app_id = "quest.cmu.twa" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "kechezon@andrew.cmu.edu" } # googleplay_app_iam.quest_xboxbedrock will be created + resource "googleplay_app_iam" "quest_xboxbedrock" { + app_id = "quest.cmu.twa" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_APP_CONTENT", + "CAN_MANAGE_PUBLIC_APKS", + "CAN_MANAGE_PUBLIC_LISTING", + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_REPLY_TO_REVIEWS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "spola2@andrew.cmu.edu" } # googleplay_app_iam.terrier_dkou will be created + resource "googleplay_app_iam" "terrier_dkou" { + app_id = "build.terrier.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "dkou@andrew.cmu.edu" } # googleplay_app_iam.terrier_kothurvidur will be created + resource "googleplay_app_iam" "terrier_kothurvidur" { + app_id = "build.terrier.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "vkothur@andrew.cmu.edu" } Plan: 5 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 5 to add, 0 to change, 0 to destroy. --- * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Author
Owner

atlantis apply

atlantis apply

Ran Apply for project: google_play dir: tofu/google_play workspace: default

Apply Error

Show Output
running 'sh -c' '/nix/store/9dj5sdaq6vksi96ibn9vpyzjcbhf26jv-opentofu-1.12.5/bin/tofu apply -input=false "/var/lib/atlantis/repos/ScottyLabs/governance/178/default/tofu/google_play/google_play-default.tfplan"' in '/var/lib/atlantis/repos/ScottyLabs/governance/178/default/tofu/google_play': exit status 1
    Finished `release` profile [optimized] target(s) in 0.05s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
googleplay_app_iam.quest_kechezon: Creating...
googleplay_app_iam.terrier_dkou: Creating...
googleplay_app_iam.terrier_kothurvidur: Creating...
googleplay_app_iam.quest_brightz: Creating...
googleplay_app_iam.quest_xboxbedrock: Creating...
╷
│ Error: Failed to grant access to app:
│ 
│   with googleplay_app_iam.quest_brightz,
│   on google_play.tf.json line 13, in resource.googleplay_app_iam.quest_brightz:
│   13:       },
│ 
│ googleapi: Error 404: The user
│ developers/7984360456919420393/users/brightz@andrew.cmu.edu does not exist,
│ notFound
╵
╷
│ Error: Failed to grant access to app:
│ 
│   with googleplay_app_iam.quest_kechezon,
│   on google_play.tf.json line 23, in resource.googleplay_app_iam.quest_kechezon:
│   23:       },
│ 
│ googleapi: Error 404: The user
│ developers/7984360456919420393/users/kechezon@andrew.cmu.edu does not
│ exist, notFound
╵
╷
│ Error: Failed to grant access to app:
│ 
│   with googleplay_app_iam.quest_xboxbedrock,
│   on google_play.tf.json line 37, in resource.googleplay_app_iam.quest_xboxbedrock:
│   37:       },
│ 
│ googleapi: Error 404: The user
│ developers/7984360456919420393/users/spola2@andrew.cmu.edu does not exist,
│ notFound
╵
╷
│ Error: Failed to grant access to app:
│ 
│   with googleplay_app_iam.terrier_dkou,
│   on google_play.tf.json line 47, in resource.googleplay_app_iam.terrier_dkou:
│   47:       },
│ 
│ googleapi: Error 404: The user
│ developers/7984360456919420393/users/dkou@andrew.cmu.edu does not exist,
│ notFound
╵
╷
│ Error: Failed to grant access to app:
│ 
│   with googleplay_app_iam.terrier_kothurvidur,
│   on google_play.tf.json line 57, in resource.googleplay_app_iam.terrier_kothurvidur:
│   57:       }
│ 
│ googleapi: Error 404: The user
│ developers/7984360456919420393/users/vkothur@andrew.cmu.edu does not exist,
│ notFound
╵

Ran Apply for project: `google_play` dir: `tofu/google_play` workspace: `default` **Apply Error** <details><summary>Show Output</summary> ``` running 'sh -c' '/nix/store/9dj5sdaq6vksi96ibn9vpyzjcbhf26jv-opentofu-1.12.5/bin/tofu apply -input=false "/var/lib/atlantis/repos/ScottyLabs/governance/178/default/tofu/google_play/google_play-default.tfplan"' in '/var/lib/atlantis/repos/ScottyLabs/governance/178/default/tofu/google_play': exit status 1 Finished `release` profile [optimized] target(s) in 0.05s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ googleplay_app_iam.quest_kechezon: Creating... googleplay_app_iam.terrier_dkou: Creating... googleplay_app_iam.terrier_kothurvidur: Creating... googleplay_app_iam.quest_brightz: Creating... googleplay_app_iam.quest_xboxbedrock: Creating... ╷ │ Error: Failed to grant access to app: │ │ with googleplay_app_iam.quest_brightz, │ on google_play.tf.json line 13, in resource.googleplay_app_iam.quest_brightz: │ 13: }, │ │ googleapi: Error 404: The user │ developers/7984360456919420393/users/brightz@andrew.cmu.edu does not exist, │ notFound ╵ ╷ │ Error: Failed to grant access to app: │ │ with googleplay_app_iam.quest_kechezon, │ on google_play.tf.json line 23, in resource.googleplay_app_iam.quest_kechezon: │ 23: }, │ │ googleapi: Error 404: The user │ developers/7984360456919420393/users/kechezon@andrew.cmu.edu does not │ exist, notFound ╵ ╷ │ Error: Failed to grant access to app: │ │ with googleplay_app_iam.quest_xboxbedrock, │ on google_play.tf.json line 37, in resource.googleplay_app_iam.quest_xboxbedrock: │ 37: }, │ │ googleapi: Error 404: The user │ developers/7984360456919420393/users/spola2@andrew.cmu.edu does not exist, │ notFound ╵ ╷ │ Error: Failed to grant access to app: │ │ with googleplay_app_iam.terrier_dkou, │ on google_play.tf.json line 47, in resource.googleplay_app_iam.terrier_dkou: │ 47: }, │ │ googleapi: Error 404: The user │ developers/7984360456919420393/users/dkou@andrew.cmu.edu does not exist, │ notFound ╵ ╷ │ Error: Failed to grant access to app: │ │ with googleplay_app_iam.terrier_kothurvidur, │ on google_play.tf.json line 57, in resource.googleplay_app_iam.terrier_kothurvidur: │ 57: } │ │ googleapi: Error 404: The user │ developers/7984360456919420393/users/vkothur@andrew.cmu.edu does not exist, │ notFound ╵ ``` </details>
docs: update to use package name
All checks were successful
atlantis/apply 9/9 projects applied successfully.
CI / build (pull_request) Has been skipped
atlantis/plan: slack No changes. Your infrastructure matches the configuration.
atlantis/plan: openbao No changes. Your infrastructure matches the configuration.
atlantis/plan: posthog No changes. Your infrastructure matches the configuration.
atlantis/plan: matrix_bridges No changes. Your infrastructure matches the configuration.
atlantis/plan: sentry No changes. Your infrastructure matches the configuration.
atlantis/plan: discord No changes. Your infrastructure matches the configuration.
atlantis/plan: keycloak No changes. Your infrastructure matches the configuration.
kennel/build skipped: no services or static sites defined
atlantis/plan: git No changes. Your infrastructure matches the configuration.
atlantis/plan: google_play Plan: 10 to add, 0 to change, 0 to destroy.
atlantis/plan 9/9 projects planned successfully.
atlantis/policy_check 0/0 projects had policies checked successfully.
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/apply: sentry Apply succeeded.
atlantis/apply: slack Apply succeeded.
atlantis/apply: matrix_bridges Apply succeeded.
atlantis/apply: discord Apply succeeded.
atlantis/apply: git Apply succeeded.
atlantis/apply: openbao Apply succeeded.
atlantis/apply: posthog Apply succeeded.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply: google_play Apply succeeded.
Validate / Governance (pull_request_target) Successful in 4m30s
CI / check-1 (pull_request) Successful in 5m54s
CI / check (pull_request) Successful in 0s
9241e2e6ce
Signed-off-by: Anish Pallati <i@anish.land>

Ran Plan for 9 projects:

  1. project: slack dir: tofu/slack workspace: default
  2. project: google_play dir: tofu/google_play workspace: default
  3. project: openbao dir: tofu/openbao workspace: default
  4. project: posthog dir: tofu/posthog workspace: default
  5. project: matrix_bridges dir: tofu/matrix_bridges workspace: default
  6. project: sentry dir: tofu/sentry workspace: default
  7. project: discord dir: tofu/discord workspace: default
  8. project: keycloak dir: tofu/keycloak workspace: default
  9. project: git dir: tofu/git workspace: default

1. project: slack dir: tofu/slack workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.34s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p slack
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p slack
    

2. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.37s

data.external.identity_thesuperrl: Read complete after 0s [id=-]
data.external.identity_riztofu: Read complete after 0s [id=-]

OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # googleplay_app_iam.quest_brightz will be created
+ resource "googleplay_app_iam" "quest_brightz" {
+       app_id               = "quest.cmu.twa"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "brightz@andrew.cmu.edu"
    }

  # googleplay_app_iam.quest_kechezon will be created
+ resource "googleplay_app_iam" "quest_kechezon" {
+       app_id               = "quest.cmu.twa"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "kechezon@andrew.cmu.edu"
    }

  # googleplay_app_iam.quest_xboxbedrock will be created
+ resource "googleplay_app_iam" "quest_xboxbedrock" {
+       app_id               = "quest.cmu.twa"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_APP_CONTENT",
+           "CAN_MANAGE_PUBLIC_APKS",
+           "CAN_MANAGE_PUBLIC_LISTING",
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_REPLY_TO_REVIEWS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "spola2@andrew.cmu.edu"
    }

  # googleplay_app_iam.terrier_dkou will be created
+ resource "googleplay_app_iam" "terrier_dkou" {
+       app_id               = "build.terrier.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "dkou@andrew.cmu.edu"
    }

  # googleplay_app_iam.terrier_kothurvidur will be created
+ resource "googleplay_app_iam" "terrier_kothurvidur" {
+       app_id               = "build.terrier.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "vkothur@andrew.cmu.edu"
    }

Plan: 5 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 5 to add, 0 to change, 0 to destroy.


3. project: openbao dir: tofu/openbao workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.39s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p openbao
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p openbao
    

4. project: posthog dir: tofu/posthog workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.30s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p posthog
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p posthog
    

5. project: matrix_bridges dir: tofu/matrix_bridges workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.37s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p matrix_bridges
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p matrix_bridges
    

6. project: sentry dir: tofu/sentry workspace: default

Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.13s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p sentry
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p sentry
    

7. project: discord dir: tofu/discord workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.36s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p discord
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p discord
    

8. project: keycloak dir: tofu/keycloak workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.28s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

9. project: git dir: tofu/git workspace: default

Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.31s


No changes. Your infrastructure matches the configuration.

OpenTofu has compared your real infrastructure against your configuration and
found no differences, so no changes are needed.
  • ▶️ To apply this plan, comment:
    atlantis apply -p git
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p git
    

Plan Summary

9 projects, 1 with changes, 8 with no changes, 0 failed

  • To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for 9 projects: 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` 1. project: `openbao` dir: `tofu/openbao` workspace: `default` 1. project: `posthog` dir: `tofu/posthog` workspace: `default` 1. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` 1. project: `sentry` dir: `tofu/sentry` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` --- ### 1. project: `slack` dir: `tofu/slack` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.34s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p slack ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fslack%252Fdefault%252Fslack) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p slack ``` --- ### 2. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.37s data.external.identity_thesuperrl: Read complete after 0s [id=-] data.external.identity_riztofu: Read complete after 0s [id=-] OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # googleplay_app_iam.quest_brightz will be created + resource "googleplay_app_iam" "quest_brightz" { + app_id = "quest.cmu.twa" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "brightz@andrew.cmu.edu" } # googleplay_app_iam.quest_kechezon will be created + resource "googleplay_app_iam" "quest_kechezon" { + app_id = "quest.cmu.twa" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "kechezon@andrew.cmu.edu" } # googleplay_app_iam.quest_xboxbedrock will be created + resource "googleplay_app_iam" "quest_xboxbedrock" { + app_id = "quest.cmu.twa" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_APP_CONTENT", + "CAN_MANAGE_PUBLIC_APKS", + "CAN_MANAGE_PUBLIC_LISTING", + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_REPLY_TO_REVIEWS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "spola2@andrew.cmu.edu" } # googleplay_app_iam.terrier_dkou will be created + resource "googleplay_app_iam" "terrier_dkou" { + app_id = "build.terrier.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "dkou@andrew.cmu.edu" } # googleplay_app_iam.terrier_kothurvidur will be created + resource "googleplay_app_iam" "terrier_kothurvidur" { + app_id = "build.terrier.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "vkothur@andrew.cmu.edu" } Plan: 5 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 5 to add, 0 to change, 0 to destroy. --- ### 3. project: `openbao` dir: `tofu/openbao` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.39s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p openbao ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fopenbao%252Fdefault%252Fopenbao) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p openbao ``` --- ### 4. project: `posthog` dir: `tofu/posthog` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.30s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p posthog ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fposthog%252Fdefault%252Fposthog) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p posthog ``` --- ### 5. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.37s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p matrix_bridges ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fmatrix_bridges%252Fdefault%252Fmatrix_bridges) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p matrix_bridges ``` --- ### 6. project: `sentry` dir: `tofu/sentry` workspace: `default` ```diff Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.13s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p sentry ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fsentry%252Fdefault%252Fsentry) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p sentry ``` --- ### 7. project: `discord` dir: `tofu/discord` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.36s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p discord ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fdiscord%252Fdefault%252Fdiscord) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p discord ``` --- ### 8. project: `keycloak` dir: `tofu/keycloak` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.28s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` --- ### 9. project: `git` dir: `tofu/git` workspace: `default` ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.31s No changes. Your infrastructure matches the configuration. OpenTofu has compared your real infrastructure against your configuration and found no differences, so no changes are needed. ``` * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p git ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgit%252Fdefault%252Fgit) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p git ``` --- ### Plan Summary 9 projects, 1 with changes, 8 with no changes, 0 failed * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Author
Owner

atlantis plan -p google_play

atlantis plan -p google_play

Ran Plan for project: google_play dir: tofu/google_play workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.05s

data.external.identity_kothurvidur: Read complete after 0s [id=-]
data.external.identity_xboxbedrock: Read complete after 0s [id=-]
data.external.identity_thesuperrl: Read complete after 0s [id=-]
data.external.identity_jiyas: Read complete after 0s [id=-]
data.external.identity_rzhu3: Read complete after 0s [id=-]
data.external.identity_jc_965: Read complete after 0s [id=-]
data.external.identity_liamwalk: Read complete after 0s [id=-]
data.external.identity_dkou: Read complete after 0s [id=-]
data.external.identity_brightz: Read complete after 0s [id=-]

OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # googleplay_app_iam.quest_brightz will be created
+ resource "googleplay_app_iam" "quest_brightz" {
+       app_id               = "quest.cmu.twa"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "brightz@andrew.cmu.edu"
    }

  # googleplay_app_iam.quest_kechezon will be created
+ resource "googleplay_app_iam" "quest_kechezon" {
+       app_id               = "quest.cmu.twa"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "kechezon@andrew.cmu.edu"
    }

  # googleplay_app_iam.quest_xboxbedrock will be created
+ resource "googleplay_app_iam" "quest_xboxbedrock" {
+       app_id               = "quest.cmu.twa"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_APP_CONTENT",
+           "CAN_MANAGE_PUBLIC_APKS",
+           "CAN_MANAGE_PUBLIC_LISTING",
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_REPLY_TO_REVIEWS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "spola2@andrew.cmu.edu"
    }

  # googleplay_app_iam.terrier_dkou will be created
+ resource "googleplay_app_iam" "terrier_dkou" {
+       app_id               = "build.terrier.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "dkou@andrew.cmu.edu"
    }

  # googleplay_app_iam.terrier_kothurvidur will be created
+ resource "googleplay_app_iam" "terrier_kothurvidur" {
+       app_id               = "build.terrier.app"
+       expanded_permissions = (known after apply)
+       permissions          = [
+           "CAN_MANAGE_TRACK_APKS",
+           "CAN_MANAGE_TRACK_USERS",
+           "CAN_VIEW_APP_QUALITY",
+           "CAN_VIEW_NON_FINANCIAL_DATA",
        ]
+       user_id              = "vkothur@andrew.cmu.edu"
    }

  # googleplay_user.brightz will be created
+ resource "googleplay_user" "brightz" {
+       email                = "brightz@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

  # googleplay_user.dkou will be created
+ resource "googleplay_user" "dkou" {
+       email                = "dkou@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

  # googleplay_user.kechezon will be created
+ resource "googleplay_user" "kechezon" {
+       email                = "kechezon@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

  # googleplay_user.kothurvidur will be created
+ resource "googleplay_user" "kothurvidur" {
+       email                = "vkothur@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

  # googleplay_user.xboxbedrock will be created
+ resource "googleplay_user" "xboxbedrock" {
+       email                = "spola2@andrew.cmu.edu"
+       expanded_permissions = (known after apply)
+       global_permissions   = [
+           "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL",
        ]
+       id                   = (known after apply)
+       name                 = (known after apply)
    }

Plan: 10 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p google_play
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p google_play
    

Plan: 10 to add, 0 to change, 0 to destroy.


  • To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.05s data.external.identity_kothurvidur: Read complete after 0s [id=-] data.external.identity_xboxbedrock: Read complete after 0s [id=-] data.external.identity_thesuperrl: Read complete after 0s [id=-] data.external.identity_jiyas: Read complete after 0s [id=-] data.external.identity_rzhu3: Read complete after 0s [id=-] data.external.identity_jc_965: Read complete after 0s [id=-] data.external.identity_liamwalk: Read complete after 0s [id=-] data.external.identity_dkou: Read complete after 0s [id=-] data.external.identity_brightz: Read complete after 0s [id=-] OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # googleplay_app_iam.quest_brightz will be created + resource "googleplay_app_iam" "quest_brightz" { + app_id = "quest.cmu.twa" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "brightz@andrew.cmu.edu" } # googleplay_app_iam.quest_kechezon will be created + resource "googleplay_app_iam" "quest_kechezon" { + app_id = "quest.cmu.twa" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "kechezon@andrew.cmu.edu" } # googleplay_app_iam.quest_xboxbedrock will be created + resource "googleplay_app_iam" "quest_xboxbedrock" { + app_id = "quest.cmu.twa" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_APP_CONTENT", + "CAN_MANAGE_PUBLIC_APKS", + "CAN_MANAGE_PUBLIC_LISTING", + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_REPLY_TO_REVIEWS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "spola2@andrew.cmu.edu" } # googleplay_app_iam.terrier_dkou will be created + resource "googleplay_app_iam" "terrier_dkou" { + app_id = "build.terrier.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "dkou@andrew.cmu.edu" } # googleplay_app_iam.terrier_kothurvidur will be created + resource "googleplay_app_iam" "terrier_kothurvidur" { + app_id = "build.terrier.app" + expanded_permissions = (known after apply) + permissions = [ + "CAN_MANAGE_TRACK_APKS", + "CAN_MANAGE_TRACK_USERS", + "CAN_VIEW_APP_QUALITY", + "CAN_VIEW_NON_FINANCIAL_DATA", ] + user_id = "vkothur@andrew.cmu.edu" } # googleplay_user.brightz will be created + resource "googleplay_user" "brightz" { + email = "brightz@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } # googleplay_user.dkou will be created + resource "googleplay_user" "dkou" { + email = "dkou@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } # googleplay_user.kechezon will be created + resource "googleplay_user" "kechezon" { + email = "kechezon@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } # googleplay_user.kothurvidur will be created + resource "googleplay_user" "kothurvidur" { + email = "vkothur@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } # googleplay_user.xboxbedrock will be created + resource "googleplay_user" "xboxbedrock" { + email = "spola2@andrew.cmu.edu" + expanded_permissions = (known after apply) + global_permissions = [ + "CAN_VIEW_NON_FINANCIAL_DATA_GLOBAL", ] + id = (known after apply) + name = (known after apply) } Plan: 10 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p google_play ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fgoogle_play%252Fdefault%252Fgoogle_play) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p google_play ``` Plan: 10 to add, 0 to change, 0 to destroy. --- * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
Author
Owner

atlantis apply

atlantis apply

Ran Apply for 9 projects:

  1. project: sentry dir: tofu/sentry workspace: default
  2. project: slack dir: tofu/slack workspace: default
  3. project: matrix_bridges dir: tofu/matrix_bridges workspace: default
  4. project: discord dir: tofu/discord workspace: default
  5. project: git dir: tofu/git workspace: default
  6. project: openbao dir: tofu/openbao workspace: default
  7. project: posthog dir: tofu/posthog workspace: default
  8. project: keycloak dir: tofu/keycloak workspace: default
  9. project: google_play dir: tofu/google_play workspace: default

1. project: sentry dir: tofu/sentry workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.10s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

2. project: slack dir: tofu/slack workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.32s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

3. project: matrix_bridges dir: tofu/matrix_bridges workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.31s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

4. project: discord dir: tofu/discord workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.44s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

5. project: git dir: tofu/git workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.18s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

6. project: openbao dir: tofu/openbao workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.41s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

7. project: posthog dir: tofu/posthog workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Finished `release` profile [optimized] target(s) in 0.28s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

8. project: keycloak dir: tofu/keycloak workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.37s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵

Apply complete! Resources: 0 added, 0 changed, 0 destroyed.

9. project: google_play dir: tofu/google_play workspace: default

Show Output
Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on package cache
    Blocking waiting for file lock on build directory
    Finished `release` profile [optimized] target(s) in 0.34s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
googleplay_user.kechezon: Creating...
googleplay_user.xboxbedrock: Creating...
googleplay_user.dkou: Creating...
googleplay_user.brightz: Creating...
googleplay_user.kothurvidur: Creating...
googleplay_user.brightz: Creation complete after 1s [id=brightz@andrew.cmu.edu]
googleplay_app_iam.quest_brightz: Creating...
googleplay_user.kechezon: Creation complete after 1s [id=kechezon@andrew.cmu.edu]
googleplay_app_iam.quest_kechezon: Creating...
googleplay_user.xboxbedrock: Creation complete after 1s [id=spola2@andrew.cmu.edu]
googleplay_app_iam.quest_xboxbedrock: Creating...
googleplay_user.dkou: Creation complete after 1s [id=dkou@andrew.cmu.edu]
googleplay_app_iam.terrier_dkou: Creating...
googleplay_user.kothurvidur: Creation complete after 1s [id=vkothur@andrew.cmu.edu]
googleplay_app_iam.terrier_kothurvidur: Creating...
googleplay_app_iam.terrier_dkou: Creation complete after 0s
googleplay_app_iam.quest_xboxbedrock: Creation complete after 0s
googleplay_app_iam.quest_brightz: Creation complete after 1s
googleplay_app_iam.quest_kechezon: Creation complete after 1s
googleplay_app_iam.terrier_kothurvidur: Creation complete after 1s

Apply complete! Resources: 10 added, 0 changed, 0 destroyed.

Apply Summary

9 projects, 9 successful, 0 failed, 0 errored

Ran Apply for 9 projects: 1. project: `sentry` dir: `tofu/sentry` workspace: `default` 1. project: `slack` dir: `tofu/slack` workspace: `default` 1. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` 1. project: `discord` dir: `tofu/discord` workspace: `default` 1. project: `git` dir: `tofu/git` workspace: `default` 1. project: `openbao` dir: `tofu/openbao` workspace: `default` 1. project: `posthog` dir: `tofu/posthog` workspace: `default` 1. project: `keycloak` dir: `tofu/keycloak` workspace: `default` 1. project: `google_play` dir: `tofu/google_play` workspace: `default` --- ### 1. project: `sentry` dir: `tofu/sentry` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.10s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 2. project: `slack` dir: `tofu/slack` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.32s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 3. project: `matrix_bridges` dir: `tofu/matrix_bridges` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.31s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 4. project: `discord` dir: `tofu/discord` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.44s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 5. project: `git` dir: `tofu/git` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.18s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 6. project: `openbao` dir: `tofu/openbao` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.41s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 7. project: `posthog` dir: `tofu/posthog` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Finished `release` profile [optimized] target(s) in 0.28s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 8. project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.37s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ Apply complete! Resources: 0 added, 0 changed, 0 destroyed. ``` </details> --- ### 9. project: `google_play` dir: `tofu/google_play` workspace: `default` <details><summary>Show Output</summary> ```diff Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on package cache Blocking waiting for file lock on build directory Finished `release` profile [optimized] target(s) in 0.34s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ googleplay_user.kechezon: Creating... googleplay_user.xboxbedrock: Creating... googleplay_user.dkou: Creating... googleplay_user.brightz: Creating... googleplay_user.kothurvidur: Creating... googleplay_user.brightz: Creation complete after 1s [id=brightz@andrew.cmu.edu] googleplay_app_iam.quest_brightz: Creating... googleplay_user.kechezon: Creation complete after 1s [id=kechezon@andrew.cmu.edu] googleplay_app_iam.quest_kechezon: Creating... googleplay_user.xboxbedrock: Creation complete after 1s [id=spola2@andrew.cmu.edu] googleplay_app_iam.quest_xboxbedrock: Creating... googleplay_user.dkou: Creation complete after 1s [id=dkou@andrew.cmu.edu] googleplay_app_iam.terrier_dkou: Creating... googleplay_user.kothurvidur: Creation complete after 1s [id=vkothur@andrew.cmu.edu] googleplay_app_iam.terrier_kothurvidur: Creating... googleplay_app_iam.terrier_dkou: Creation complete after 0s googleplay_app_iam.quest_xboxbedrock: Creation complete after 0s googleplay_app_iam.quest_brightz: Creation complete after 1s googleplay_app_iam.quest_kechezon: Creation complete after 1s googleplay_app_iam.terrier_kothurvidur: Creation complete after 1s Apply complete! Resources: 10 added, 0 changed, 0 destroyed. ``` </details> --- ### Apply Summary 9 projects, 9 successful, 0 failed, 0 errored
anish merged commit 24b3043ae3 into main 2026-08-24 21:24:29 +00:00
anish deleted branch google-play 2026-08-24 21:24:30 +00:00

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/discord workspace: default
  • dir: tofu/git workspace: default
  • dir: tofu/google_play workspace: default
  • dir: tofu/keycloak workspace: default
  • dir: tofu/matrix_bridges workspace: default
  • dir: tofu/openbao workspace: default
  • dir: tofu/posthog workspace: default
  • dir: tofu/sentry workspace: default
  • dir: tofu/slack workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/discord` workspace: `default` - dir: `tofu/git` workspace: `default` - dir: `tofu/google_play` workspace: `default` - dir: `tofu/keycloak` workspace: `default` - dir: `tofu/matrix_bridges` workspace: `default` - dir: `tofu/openbao` workspace: `default` - dir: `tofu/posthog` workspace: `default` - dir: `tofu/sentry` workspace: `default` - dir: `tofu/slack` workspace: `default`
Sign in to join this conversation.
No description provided.