feat: add custom groups to oquest #166

Merged
anish merged 2 commits from add-quest-groups into main 2026-08-10 23:27:22 +00:00
Member
No description provided.
feat: add custom groups to oquest
All checks were successful
CI / build (pull_request) Has been skipped
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
kennel/build skipped: no services or static sites defined
Validate / Governance (pull_request_target) Successful in 13m19s
CI / check-1 (pull_request) Successful in 23m56s
CI / check (pull_request) Successful in 0s
e889e4a6fb
chore: regenerate tofu, schemas, CODEOWNERS [skip ci]
All checks were successful
atlantis/plan: keycloak Plan: 3 to add, 0 to change, 0 to destroy.
kennel/build skipped: no services or static sites defined
atlantis/pre_workflow_hook: Install local-only OpenTofu providers succeeded.
atlantis/plan 1/1 projects planned successfully.
atlantis/apply: keycloak Apply succeeded.
atlantis/apply 1/1 projects applied successfully.
441ed5a0b8

Ran Plan for project: keycloak dir: tofu/keycloak workspace: default

Show Output
Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/166/default/crates/governance-schema)
   Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/166/default/crates/governance-core)
   Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/166/default/crates/governance-tfgen)
   Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/166/default/crates/governance-cli)
    Finished `release` profile [optimized] target(s) in 7.75s


OpenTofu used the selected providers to generate the following execution
plan. Resource actions are indicated with the following symbols:
+ create

OpenTofu will perform the following actions:

  # keycloak_group.project_quest_group_challenge-placer will be created
+ resource "keycloak_group" "project_quest_group_challenge-placer" {
+       id        = (known after apply)
+       name      = "challenge-placer"
+       parent_id = "cd488cc4-cbe6-4bdd-9a96-e1358077a371"
+       path      = (known after apply)
+       realm_id  = "scottylabs"
    }

  # keycloak_group.project_quest_group_orientation-staff will be created
+ resource "keycloak_group" "project_quest_group_orientation-staff" {
+       id        = (known after apply)
+       name      = "orientation-staff"
+       parent_id = "cd488cc4-cbe6-4bdd-9a96-e1358077a371"
+       path      = (known after apply)
+       realm_id  = "scottylabs"
    }

  # keycloak_group.project_quest_group_trade-admin will be created
+ resource "keycloak_group" "project_quest_group_trade-admin" {
+       id        = (known after apply)
+       name      = "trade-admin"
+       parent_id = "cd488cc4-cbe6-4bdd-9a96-e1358077a371"
+       path      = (known after apply)
+       realm_id  = "scottylabs"
    }

Plan: 3 to add, 0 to change, 0 to destroy.
  • ▶️ To apply this plan, comment:
    atlantis apply -p keycloak
    
  • 🚮 To delete this plan and lock, click here
  • 🔁 To plan this project again, comment:
    atlantis plan -p keycloak
    

Plan: 3 to add, 0 to change, 0 to destroy.


  • To apply all unapplied plans from this Pull Request, comment:
    atlantis apply
    
  • 🚮 To delete all plans and locks from this Pull Request, comment:
    atlantis unlock
    
Ran Plan for project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Compiling governance-schema v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/166/default/crates/governance-schema) Compiling governance-core v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/166/default/crates/governance-core) Compiling governance-tfgen v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/166/default/crates/governance-tfgen) Compiling governance v0.1.0 (/var/lib/atlantis/repos/ScottyLabs/governance/166/default/crates/governance-cli) Finished `release` profile [optimized] target(s) in 7.75s OpenTofu used the selected providers to generate the following execution plan. Resource actions are indicated with the following symbols: + create OpenTofu will perform the following actions: # keycloak_group.project_quest_group_challenge-placer will be created + resource "keycloak_group" "project_quest_group_challenge-placer" { + id = (known after apply) + name = "challenge-placer" + parent_id = "cd488cc4-cbe6-4bdd-9a96-e1358077a371" + path = (known after apply) + realm_id = "scottylabs" } # keycloak_group.project_quest_group_orientation-staff will be created + resource "keycloak_group" "project_quest_group_orientation-staff" { + id = (known after apply) + name = "orientation-staff" + parent_id = "cd488cc4-cbe6-4bdd-9a96-e1358077a371" + path = (known after apply) + realm_id = "scottylabs" } # keycloak_group.project_quest_group_trade-admin will be created + resource "keycloak_group" "project_quest_group_trade-admin" { + id = (known after apply) + name = "trade-admin" + parent_id = "cd488cc4-cbe6-4bdd-9a96-e1358077a371" + path = (known after apply) + realm_id = "scottylabs" } Plan: 3 to add, 0 to change, 0 to destroy. ``` </details> * :arrow_forward: To **apply** this plan, comment: ```shell atlantis apply -p keycloak ``` * :put_litter_in_its_place: To **delete** this plan and lock, click [here](https://atlantis.scottylabs.org/lock?id=ScottyLabs%252Fgovernance%252Ftofu%252Fkeycloak%252Fdefault%252Fkeycloak) * :repeat: To **plan** this project again, comment: ```shell atlantis plan -p keycloak ``` Plan: 3 to add, 0 to change, 0 to destroy. --- * :fast_forward: To **apply** all unapplied plans from this Pull Request, comment: ```shell atlantis apply ``` * :put_litter_in_its_place: To **delete** all plans and locks from this Pull Request, comment: ```shell atlantis unlock ```
anish approved these changes 2026-08-10 23:26:57 +00:00
Owner

atlantis apply

atlantis apply

Ran Apply for project: keycloak dir: tofu/keycloak workspace: default

Show Output
Finished `release` profile [optimized] target(s) in 0.05s

╷
│ Warning: Provider development overrides are in effect
│ 
│ The following provider development overrides are set in the CLI
│ configuration:
│  - svalabs/forgejo in /tmp/forgejo-provider
│ 
│ The behavior may therefore not match any released version of the provider
│ and applying changes may cause the state to become incompatible with
│ published releases.
╵
keycloak_group.project_quest_group_challenge-placer: Creating...
keycloak_group.project_quest_group_orientation-staff: Creating...
keycloak_group.project_quest_group_trade-admin: Creating...
keycloak_group.project_quest_group_challenge-placer: Creation complete after 0s [id=8ee630ad-174b-4f64-b923-5c5dc1941498]
keycloak_group.project_quest_group_orientation-staff: Creation complete after 0s [id=08760b29-aa23-4d1f-87ab-ba4608c28970]
keycloak_group.project_quest_group_trade-admin: Creation complete after 0s [id=2295dbb5-6321-45b3-b00e-1c91b6819bbe]

Apply complete! Resources: 3 added, 0 changed, 0 destroyed.
Ran Apply for project: `keycloak` dir: `tofu/keycloak` workspace: `default` <details><summary>Show Output</summary> ```diff Finished `release` profile [optimized] target(s) in 0.05s ╷ │ Warning: Provider development overrides are in effect │ │ The following provider development overrides are set in the CLI │ configuration: │ - svalabs/forgejo in /tmp/forgejo-provider │ │ The behavior may therefore not match any released version of the provider │ and applying changes may cause the state to become incompatible with │ published releases. ╵ keycloak_group.project_quest_group_challenge-placer: Creating... keycloak_group.project_quest_group_orientation-staff: Creating... keycloak_group.project_quest_group_trade-admin: Creating... keycloak_group.project_quest_group_challenge-placer: Creation complete after 0s [id=8ee630ad-174b-4f64-b923-5c5dc1941498] keycloak_group.project_quest_group_orientation-staff: Creation complete after 0s [id=08760b29-aa23-4d1f-87ab-ba4608c28970] keycloak_group.project_quest_group_trade-admin: Creation complete after 0s [id=2295dbb5-6321-45b3-b00e-1c91b6819bbe] Apply complete! Resources: 3 added, 0 changed, 0 destroyed. ``` </details>
anish merged commit dc248d6f34 into main 2026-08-10 23:27:22 +00:00
anish deleted branch add-quest-groups 2026-08-10 23:27:22 +00:00

Locks and plans deleted for the projects and workspaces modified in this pull request:

  • dir: tofu/keycloak workspace: default
Locks and plans deleted for the projects and workspaces modified in this pull request: - dir: `tofu/keycloak` workspace: `default`
Sign in to join this conversation.
No description provided.