feat: repoint to git.cmu.dev #150
No reviewers
Labels
No labels
needs/linking
needs/docs
needs/research
needs/upstream
type/bug
type/external
type/feature
type/refactor
urgency
high
urgency
immediate
urgency
low
urgency
medium
No milestone
No project
No assignees
2 participants
Notifications
Due date
No due date set.
Dependencies
No dependencies set
Reference
ScottyLabs/governance!150
Loading…
Reference in a new issue
No description provided.
Delete branch "git-cmu-dev"
Deleting a branch is permanent. Although the deleted branch may continue to exist for a short time before it actually gets removed, it CANNOT be undone in most cases. Continue?
590a495d2ff5a9aa354df5a9aa354d283d7882c3283d7882c3d1f2bfe0d0d1f2bfe0d03bb6e600f53bb6e600f5c2256a562bc2256a562bd77dbf965ad77dbf965a26585a24baRan Plan for 6 projects:
slackdir:tofu/slackworkspace:defaultvaultwardendir:tofu/vaultwardenworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
Plan: 0 to add, 0 to change, 77 to destroy.
2. project:
vaultwardendir:tofu/vaultwardenworkspace:defaultShow Output
Plan: 1 to add, 1 to change, 0 to destroy.
3. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
Plan: 0 to add, 0 to change, 43 to destroy.
4. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
Plan: 0 to add, 79 to change, 0 to destroy.
5. project:
discorddir:tofu/discordworkspace:defaultShow Output
Plan: 0 to add, 0 to change, 30 to destroy.
6. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 124 to add, 25 to change, 74 to destroy.
Plan Summary
6 projects, 6 with changes, 0 with no changes, 0 failed
atlantis apply -p discord -p slack -p keycloak -p google_groups -p vaultwarden
Ran Apply for project:
vaultwardendir:tofu/vaultwardenworkspace:defaultShow Output
atlantis apply -p discord
atlantis apply -p slack
Apply Error
atlantis apply -p google_groups
Apply Error
Ran Apply for project:
discorddir:tofu/discordworkspace:defaultShow Output
atlantis apply -p slack
Ran Apply for project:
slackdir:tofu/slackworkspace:defaultShow Output
atlantis apply -p google_groups
Ran Apply for project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
atlantis apply -p keycloak
Ran Apply for project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
atlantis plan
Ran Plan for 6 projects:
slackdir:tofu/slackworkspace:defaultvaultwardendir:tofu/vaultwardenworkspace:defaultdiscorddir:tofu/discordworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
No changes. Your infrastructure matches the configuration.
2. project:
vaultwardendir:tofu/vaultwardenworkspace:defaultShow Output
Plan: 0 to add, 2 to change, 0 to destroy.
3. project:
discorddir:tofu/discordworkspace:default4. project:
google_groupsdir:tofu/google_groupsworkspace:default5. project:
keycloakdir:tofu/keycloakworkspace:default6. project:
gitdir:tofu/gitworkspace:defaultShow Output
Plan: 111 to add, 25 to change, 74 to destroy.
Plan Summary
6 projects, 2 with changes, 4 with no changes, 0 failed
atlantis apply
Ran Apply for 6 projects:
slackdir:tofu/slackworkspace:defaultdiscorddir:tofu/discordworkspace:defaultkeycloakdir:tofu/keycloakworkspace:defaultgoogle_groupsdir:tofu/google_groupsworkspace:defaultvaultwardendir:tofu/vaultwardenworkspace:defaultgitdir:tofu/gitworkspace:default1. project:
slackdir:tofu/slackworkspace:defaultShow Output
2. project:
discorddir:tofu/discordworkspace:defaultShow Output
3. project:
keycloakdir:tofu/keycloakworkspace:defaultShow Output
4. project:
google_groupsdir:tofu/google_groupsworkspace:defaultShow Output
5. project:
vaultwardendir:tofu/vaultwardenworkspace:defaultShow Output
6. project:
gitdir:tofu/gitworkspace:defaultShow Output
Apply Summary
6 projects, 6 successful, 0 failed, 0 errored
Locks and plans deleted for the projects and workspaces modified in this pull request:
tofu/discordworkspace:defaulttofu/gitworkspace:defaulttofu/google_groupsworkspace:defaulttofu/keycloakworkspace:defaulttofu/slackworkspace:defaulttofu/vaultwardenworkspace:default